)]}'
{"id":"openstack%2Fcinder~810917","triplet_id":"openstack%2Fcinder~master~I0bfb79db4f69c0705a1e0fa971376d14c232d5cb","project":"openstack/cinder","branch":"master","topic":"bug/1943682","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-02-14 16:55:13.000000000","reason":"Change was submitted"},"33431":{"account":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"last_update":"2022-01-20 20:02:37.000000000","reason":"Fábio Oliveira replied on the change"},"34201":{"account":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"last_update":"2022-02-14 16:55:13.000000000","reason":"Change was submitted"},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2022-02-14 16:55:13.000000000","reason":"Change was submitted"},"32266":{"account":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"last_update":"2022-02-14 16:55:13.000000000","reason":"Change was submitted"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-02-14 12:29:10.000000000","reason":"Gorka Eguileor replied on the change"}},"hashtags":[],"change_id":"I0bfb79db4f69c0705a1e0fa971376d14c232d5cb","subject":"[SVf] update rccg name property to metadata","status":"MERGED","created":"2021-09-24 14:05:33.000000000","updated":"2022-02-14 16:57:08.000000000","submitted":"2022-02-14 16:55:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":40,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"810917-bug/1943682","meta_rev_id":"f4b1ef7805b8426c3b28182e81459f30e0477964","_number":810917,"virtual_id_number":810917,"owner":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-14 16:55:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-14 15:18:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2022-02-14 12:29:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-14 15:18:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"}],"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-24 14:33:03.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-09-24 14:36:57.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-24 14:40:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-24 15:16:14.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-09-24 15:25:09.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-24 15:33:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-25 09:32:17.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-27 02:27:32.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-28 16:58:01.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2021-09-29 09:25:14.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-29 10:13:23.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-09-29 10:22:55.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-09-29 10:25:41.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-29 11:41:43.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-09-29 12:03:41.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-09-29 12:41:41.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-09-29 13:47:53.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-09-29 15:53:44.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-29 21:06:28.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-09-30 05:51:09.000000000","updated_by":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"reviewer":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"state":"REVIEWER"},{"updated":"2021-09-30 06:11:23.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-10-01 03:25:40.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-12-14 20:03:02.000000000","updated_by":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"reviewer":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"state":"REVIEWER"},{"updated":"2021-12-23 16:03:23.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2021-12-27 07:30:13.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-27 07:34:54.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-27 07:42:00.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-27 07:50:16.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-27 08:45:04.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2022-01-03 05:06:44.000000000","updated_by":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2022-01-03 05:06:44.000000000","updated_by":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"CC"},{"updated":"2022-01-03 05:06:44.000000000","updated_by":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"CC"},{"updated":"2022-01-06 16:49:10.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2022-01-06 17:06:39.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2022-01-10 08:15:25.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2022-01-15 06:50:30.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-21 15:08:59.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2022-02-07 13:30:12.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-07 14:09:24.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-02-07 16:05:04.000000000","updated_by":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"reviewer":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"state":"CC"},{"updated":"2022-02-07 19:55:27.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-02-09 14:54:26.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"}],"messages":[{"id":"5a3496bd05ac4112d77c4a99bc4e6a7fe949d1ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"date":"2021-09-24 14:05:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c638289aa0256ecac0fab41bde0f80dfc006acf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-09-24 14:33:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/17/810917/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/fe93b24/ : RETRY_LIMIT in 7m 23s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/17/810917/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7f4650f/ : RETRY_LIMIT in 7m 18s","accounts_in_message":[],"_revision_number":1},{"id":"3af79c7861b8c9a4d40d69d114684eb412db47e0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-24 14:36:57.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-810917-1/ns5_nfs : FAILURE in 18m 18s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-810917-1/ns5_iscsi : FAILURE in 18m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"d82a9235bdc13e0c2227b3d07a850ff00de073f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 14:40:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/549f9bb677c949d887e6e6e88f4ece2f : SUCCESS in 14m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/61c17c9f54fc40ac8d999eaf902737ce : SUCCESS in 26m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6ef7b860df4c8b8d7859348997d5b7f95a98f53","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-09-24 15:16:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/810917/1/iscsi/ : SUCCESS in 1h 10m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/810917/1/smb/ : SUCCESS in 1h 08m 39s","accounts_in_message":[],"_revision_number":1},{"id":"7a0a600c984701938deb449faab080d727350789","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-24 15:25:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/17/810917/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/bf864de/bf864de.html : SUCCESS in 1h 18m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d52e27bc4bf32ce965a0420fac1a7273ea9e8706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 15:33:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/547f71e7d7a5478686d4c3c189cfc2a0 : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec0ea698222c42428c3cbeca90f7e038 : SUCCESS in 10m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/487ab720d6c04cd4a147095dcbee7c54 : SUCCESS in 14m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81eaa3ff7b7f428da5bad044a32d90fd : SUCCESS in 12m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fa528284cdd4c0dbb2a0dfbc4b24989 : SUCCESS in 15m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/3325398080284c33811f84f0903f7966 : SUCCESS in 1h 02m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fb07cd73e5a84dd9a4e75aa2f9ba491f : RETRY_LIMIT in 10m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c483411bfb842df88a712502d34e515 : SUCCESS in 7m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8959352d4657476dabdd21ed9d05e93d : SUCCESS in 18m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ccdf4163922449a398162700ad0fdb85 : SUCCESS in 8m 15s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d9a1baec588f47c9931e63a62e439724 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cd937c5ae75d4089be39b18a4644ead3 : SUCCESS in 8m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9a9775396b4a48258ebbb2adf3d2faac : SUCCESS in 8m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9221c92920ae484984a5324f5fbb124e : RETRY_LIMIT in 10m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e66a88175621434989f24056922c93ed : FAILURE in 9m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/657df63ba1d84c50aa90bc64c3a205c9 : RETRY_LIMIT in 11m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/adebe22a70594306a7cf000ff539a2c5 : RETRY_LIMIT in 18m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/276fadb91f8741ad9d92b3d8a0ce30f8 : RETRY_LIMIT in 12m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/78884d40ed8848f490ceebc962b544ad : SUCCESS in 1h 16m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/46d30c636f9c4180865ea58ab19f839b : RETRY_LIMIT in 16m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7c13d241b7aa4d179b9b041c2d12c37d : RETRY_LIMIT in 13m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/17a55eda248548edb3bf84ce1ac77acc : RETRY_LIMIT in 18m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dcf86df26aca4e5f8c22fc0c1bbe6a3b : RETRY_LIMIT in 18m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b256b74eb68d451ca70c543996c20a7f : RETRY_LIMIT in 14m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b941b07eb9974399aa025caf75fcacb0 : RETRY_LIMIT in 20m 01s","accounts_in_message":[],"_revision_number":1},{"id":"05ff36c38725028a522ae42094bd46253de874d6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-24 16:32:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/17/810917/1/check/iscsi-driver-alletra-6k/f32a912/f32a912.html : SUCCESS in 2h 25m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2bb56b9dc2a57dbc8ea0989e7fa8e20336c12fd8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-24 16:49:10.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/17/810917/1/check/iscsi-driver-nimble/3627188/3627188.html : SUCCESS in 2h 42m 27s","accounts_in_message":[],"_revision_number":1},{"id":"edef1b516d6baae6bf72f7586535fa602e18f239","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-24 23:05:33.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-810917-1/ns5_nfs : SUCCESS in 4h 12m 22s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-810917-1/ns5_iscsi : SUCCESS in 4h 16m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"f2547a5e08acbfb9bebe0b02084a270efbe9b823","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-25 02:50:58.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-810917-1/ns5_nfs : SUCCESS in 4h 12m 40s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-810917-1/ns5_iscsi : SUCCESS in 4h 17m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"27435b8a70269d902856b3d253293cad330acfb8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-25 09:32:17.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/810917/810917/1/check-not-vote/ext-cinder-zuul/6181a44 : SUCCESS in 30m 32s","accounts_in_message":[],"_revision_number":1},{"id":"480599c52d82b4852de8e299d851a3259681e7f8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-09-27 02:27:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/810917/1/20210927083657/NEC-FC : SUCCESS in 2h 50m 12s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/810917/1/20210927083657/NEC-ISCSI : SUCCESS in 2h 50m 31s","accounts_in_message":[],"_revision_number":1},{"id":"5203d4dd7c699d42cf4bc64f4bfff0178a2e987d","author":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"date":"2021-09-28 15:54:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9f1aceda1cac1d2db4e5889fabcc0a987a00588d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 16:18:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a814894a0fa64fbd8209bc5fcaec6e24 : SUCCESS in 17m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe21857e892a452bacf64d3a97f7704b : SUCCESS in 21m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"215ca55b93f36fe0a974d9aee5e56db6e7cb999b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 16:38:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7042256503c540d5bc83c8cc152bf0be : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31368f87f74c4d4ca038ceffa0255621 : SUCCESS in 11m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb963ea2e8c7499daf90ecba9c42ae69 : SUCCESS in 10m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8c339e9c06b4bcaae187539af74c90d : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db28ee200270449ca4312065c19bcf41 : SUCCESS in 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/4720f4e3e8b547bfb5f11ae7bc68a4c5 : FAILURE in 27m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/00ed83a63eea414385a203515bdcf42f : FAILURE in 29m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b0f316459ff4671a2fe70ec6ad47689 : SUCCESS in 6m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9f67a8283a0544b8b3cc0b7f5e08c2fa : SUCCESS in 14m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d05b44d191064ec8bf86ed9c5febfa31 : SUCCESS in 5m 52s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1eabbd2674d74de8a206fac5b1c519ae : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1397706c416b4a33a78e389ef2c6f9b0 : SUCCESS in 6m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ba189ccb216841d0aac19bd9a607ce5d : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/55a9be89f8134ebda68f1c659e4e7d43 : FAILURE in 29m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6057a7a66b05439a9a1fdec087bf5630 : FAILURE in 6m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1c59deba7fc8412d94706106932527ff : FAILURE in 32m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bcff52ffb7fb43e8b8d636d9ed03273c : FAILURE in 37m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b8ec86c828cb4c628c0af6ee7e37dc52 : FAILURE in 33m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d284305e2d134a6181b4121a671b5b7f : FAILURE in 36m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/75513fa2e5fa40dd87aff649c2774f8f : FAILURE in 27m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a2d6f9c3e0854ac18c11f9cded91a907 : FAILURE in 28m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c987f21aae5a4f349cbfb2e6cc743af3 : FAILURE in 30m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/625c20150fa740e2bbf7f3c93cbaa9b3 : FAILURE in 31m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f4a655888f343be8f7b2ba136b47d42 : FAILURE in 22m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a59d0d1a43c4381946131128f3a655f : POST_FAILURE in 24m 47s","accounts_in_message":[],"_revision_number":1},{"id":"7fa40f0e22f1c02645e575acc0fe82cf49dc1936","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-09-28 16:58:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/37f4443/index.html : FAILURE in 33m 17s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/dab1c32/index.html : FAILURE in 33m 56s","accounts_in_message":[],"_revision_number":1},{"id":"cc1cc186f10bb79ac5e5e5484aebe24efb6866d2","author":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"date":"2021-09-29 04:11:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2fd47bd9a1e222681daf431fc87661d410c34e28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 04:36:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7bd5ff62d85f42408349385995893eb5 : SUCCESS in 12m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/11c2c1adaee84fcb80615c42b12a4951 : SUCCESS in 23m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"88fb0d4a589e24b48f9e30d220f87fd91fad4732","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 06:24:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05ccdfc6b7f94541a325a4054663e709 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9697247c38814f1b9c7d42924d03b6dc : SUCCESS in 10m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/926a5130126e46fbada831e8242e6bb1 : SUCCESS in 11m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3699c8cf55040068ff39ba969f7ffc5 : SUCCESS in 12m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea147e148a4c41b3b14e671aabaf0900 : SUCCESS in 12m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/eb698bc6c79e461ca2059d35d16a2e87 : SUCCESS in 1h 04m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/15d5213a8b9040c3b03bd216237113d6 : SUCCESS in 1h 02m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c7a4b4edf0446e197c9023defbc81ab : SUCCESS in 7m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b9599807ee694ddcb86b8e91b0f9cd7e : SUCCESS in 14m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/327f505bfb7a40be848073103b1ab961 : SUCCESS in 6m 06s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5b804c773925454ea3e3e1f3197e3e94 : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0b96de479ee14d618df39834c62f6e41 : SUCCESS in 7m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7fa5322ca51940af87b8f58fbd54cc90 : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/29f417057f1543cb9f5562a9f58c86d6 : SUCCESS in 36m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b27958c4d3b54a80a6dee753dd137913 : FAILURE in 7m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/61c864fe485f4ebfa5ec05c1c28a80d6 : SUCCESS in 1h 14m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/36d321b5aa984d13aa1739bdb8a2dd1f : SUCCESS in 1h 30m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/692f42aea0544fa9b1cfc89f54b17942 : SUCCESS in 1h 24m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fd32f1577554454c9e6675696e38dd6a : SUCCESS in 1h 06m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/630c72706d3d45d1b268925e9f44a0c0 : SUCCESS in 1h 06m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cb33c1aee56c406a81c40ad5f733f517 : SUCCESS in 44m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4b12ca50b0a64ea6bd4c94b014dc785f : SUCCESS in 1h 09m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/26d34c39d63e4a84ae04d4d55c874237 : SUCCESS in 2h 08m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b2aa79707a647a185d9c65b876c7a18 : SUCCESS in 53m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a970c4795e8a488aac9f99b646ff8b0c : SUCCESS in 45m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b3076add38b9eb24d36e808edb30e0b4794b54af","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-09-29 06:52:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/04928f5/index.html : SUCCESS in 1h 00m 56s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/73cc411/index.html : SUCCESS in 1h 01m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1d000c7a3873e43396f3d1ef860c82eb6468fc11","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-09-29 09:25:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/17/810917/1/upstream-check/cinder-cDOT-iSCSI/db6c607/ : SUCCESS in 58m 21s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/17/810917/1/upstream-check/cinder-cDOT-NFS/e57ad33/ : SUCCESS in 36m 28s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/17/810917/1/upstream-check/cinder-cDOT-FCP/9cd72b2/ : SUCCESS in 57m 27s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d1c422506facd1c31bfaea70b039ef33e23c8a5f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-09-29 10:13:23.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/17/810917/1/check/DellEMC_Unity_FC/9ac1e8f/DellEMC_Unity_FC/978 : FAILURE in 3s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/1/check/DellEMC_Unity_iSCSI/3739f38/DellEMC_Unity_iSCSI/845 : FAILURE in 1h 29m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"e7054a01e052d6c88f2d15ba3db64fecd457ad5b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-09-29 10:22:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/17/810917/1/upstream-solidfire-check/cinder-solidfire-iSCSI/f66d363/ : FAILURE in 1h 41m 56s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"06feec9c79572f204a7b061bc7a1fdea54beb5fc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-09-29 10:25:41.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/17/810917/1/check/DellEMC_PowerFlex/45aefea/DellEMC_PowerFlex/703 : FAILURE in 1h 30m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"5656239148668b78ff570fcdad421fbaea8c80e6","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-09-29 11:41:43.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI http://os-logs.tristero.net/dell-ci/810917-1/PVME-iSCSI : SUCCESS in 42:20.03\n- PVME-FC http://os-logs.tristero.net/dell-ci/810917-1/PVME-FC : SUCCESS in 42:23.23","accounts_in_message":[],"_revision_number":1},{"id":"ec0deec3baa8cd21604d87314342cd5c453114f0","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-09-29 12:03:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/810917-1/HPMSA-iSCSI : SUCCESS in 51:54.83\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/810917-1/HPMSA-FC : SUCCESS in 31:56.82","accounts_in_message":[],"_revision_number":1},{"id":"3771c5bdd024550c82909384fe8f58e5b193b4c1","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-09-29 12:41:41.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/17/810917/1/check/DellEMC_SC_FC/3b3f9b3/DellEMC_SC_FC/620 : FAILURE in 48m 03s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/1/check/DellEMC_SC_iSCSI/11714cc/DellEMC_SC_iSCSI/588 : FAILURE in 2h 00m 49s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"5e8b28480567ee4dcf1b2b007f6f40913d93a5d5","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-09-29 13:47:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/810917-1/Lenovo-iSCSI : SUCCESS in 1:05:58\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/810917-1/Lenovo-FC : SUCCESS in 48:12.20","accounts_in_message":[],"_revision_number":1},{"id":"bfeeb66acbc6deb131323f7426b49fd0de10a891","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-09-29 15:53:44.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/1/check/DellEMC_VNX_iSCSI/c307aca/DellEMC_VNX_iSCSI/627 : FAILURE in 2h 24m 41s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/17/810917/1/check/DellEMC_VNX_FC/9f4f66f/DellEMC_VNX_FC/616 : FAILURE in 1h 39m 47s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"0240216869f11f61ee2a0901edc4fecd0c91d2cd","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-09-29 21:06:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/810917-1/Seagate-iSCSI : SUCCESS in 3:06:18\n- Seagate-FC http://os-logs.tristero.net/seagate-ci/810917-1/Seagate-FC : SUCCESS in 38:28.37","accounts_in_message":[],"_revision_number":1},{"id":"93f08d9b73607bddbff1580238ba290e951c1ed9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-30 00:27:56.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/810917/810917/1/check-not-vote/ext-cinder-zuul/c59811a : SUCCESS in 34m 24s","accounts_in_message":[],"_revision_number":1},{"id":"6bb3f96a29eb0ff112bc8e954c109a5141e0861f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-09-30 06:11:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"417cdc623d8451db629eea02e8ff2301c7129981","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-10-01 03:25:40.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/1/check/DellEMC_PowerStore_iSCSI/1804b18/DellEMC_PowerStore_iSCSI/None : RETRY_LIMIT\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/17/810917/1/check/DellEMC_PowerStore_FC/661cf48/DellEMC_PowerStore_FC/727 : FAILURE in 1h 19m 23s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"62c98a260135110ba27d4ebe6b379764994195dc","author":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"date":"2021-12-21 13:52:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c86705f4a5d73f60cd2219645f8bfbd2462912ab","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-12-23 16:03:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"db9232baadc6df05bb41457ef9881a3f3ef135dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"date":"2021-12-27 06:01:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7fc983cec560dd6af8850eb295c4d43d5b1327d8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-12-27 06:01:46.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/14767/","accounts_in_message":[],"_revision_number":2},{"id":"1e542d3030f5ad733c54bfe3684d76f7d621f58a","author":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"date":"2021-12-27 06:02:29.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6cbfac7cfa7e4b9e44673b5a8b86adc6c4674133","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-12-27 06:03:19.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/14767/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"314144d54236e369daced2ecd5c0df8fd56bdd50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 06:23:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b1d93655043486c9779402181780af1 : SUCCESS in 13m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aea4bf1348c749d79bc7df9fdf261210 : SUCCESS in 16m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8cde0f3a08eeb4d18524d8384a7cf701f7ce31ec","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-27 06:59:36.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/810917/2/iscsi/ : FAILURE in 40m 49s\n- cinder-smb http://cloudbase-ci.com/cinder/810917/2/smb/ : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":2},{"id":"8eafb951095a308209aa1b8a390d9877804cce7a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-27 07:30:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e53fddc9174042168ab2f5c1c40af111 : SUCCESS in 1h 26m 58s","accounts_in_message":[],"_revision_number":2},{"id":"f22959250d012eee14fc338182b7e838f4c35f4d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-27 07:34:54.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/17/810917/2/check/tempest-dsvm-full-focal-py3/af2832b/ : RETRY_LIMIT in 25m 32s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/17/810917/2/check/devstack-plugin-ceph-dsvm-focal-py3/efbc5d3/ : RETRY_LIMIT in 26m 10s","accounts_in_message":[],"_revision_number":2},{"id":"05ca301c79c08f625fd0d51c65aa3c4aef0fea7e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-27 07:42:00.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/17/810917/2/check/DellEMC_XtremIO_FC/dec6c0d/DellEMC_XtremIO_FC/360 : SUCCESS in 1h 05m 02s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/2/check/DellEMC_XtremIO_iSCSI/098c736/DellEMC_XtremIO_iSCSI/373 : SUCCESS in 1h 40m 01s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"40000a766ccb87b49160f5fb8d592180c6df65f1","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-12-27 07:50:16.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/17/810917/2/ : SUCCESS in 1h 40m 57s","accounts_in_message":[],"_revision_number":2},{"id":"d62e9cbdb0a8a23bb3e380461466d8090c361f14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-27 08:37:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/421743bb198f42dfb436506010fa3dcd : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15ff9f01cf104827ac1ee696c7361674 : SUCCESS in 9m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b3b3268e93f4369912438a8f9a73050 : SUCCESS in 12m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7697fc0b41354775aebe11b02359bae8 : SUCCESS in 13m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/b82ed08b3c7449cdbf775ff9586d73a7 : SUCCESS in 1h 06m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d16338b100e04d8eb937047bbfe7791e : SUCCESS in 1h 08m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7184d2b28d0149cca31422f7ad4dba2d : SUCCESS in 1h 07m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc1ae7f9aaa94033beb9a080ecb6d4aa : SUCCESS in 6m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6aa4822a55c449cf9e63321cb16dc0a7 : SUCCESS in 13m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7f74439078b2412f919faad5abd40ce7 : SUCCESS in 7m 03s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/853345fd337f4729a2f91087d06b8ac5 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b0bd8f935c204570bcad90cb1fc90e5b : SUCCESS in 6m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c551fd43909c4e018ba3fbde547d4d85 : SUCCESS in 7m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/91a9abb6d27548a1b3b9fdbe10e2509c : SUCCESS in 41m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/957fb8c8d4cd47afb226f3495f11d53c : FAILURE in 7m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/da2ea603e968430fafcaa3660276b301 : SUCCESS in 1h 03m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0bb272b14b264e74ae23d3ba15ae76a7 : SUCCESS in 1h 31m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3a7d57f26e3b42a9b0070e69cf47c66e : SUCCESS in 1h 18m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0abf92d27e2941a2a9e93528d7adc3f1 : SUCCESS in 52m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d43e5ad289f543d5be9c812e3fb4c380 : SUCCESS in 56m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c485ed56af284dd2a54bc0daacd164b7 : SUCCESS in 42m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ffcae2e3cebd4e29b5e94f3cff03e844 : SUCCESS in 1h 04m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/22d3861ae1bf4b2e920b212381a0c9f3 : SUCCESS in 2h 28m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc79ad9d137945eab840fb72a4edbd2d : SUCCESS in 55m 44s","accounts_in_message":[],"_revision_number":2},{"id":"93cb868c5de571771249fa9c53ca433090fe92b7","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-12-27 08:39:46.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"63d45127aeb3e791a8f046f08481fb642585bbbd","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-12-27 08:39:48.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"14c54b773611dcb9fac877069fc205bddba1cc00","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-12-27 08:39:48.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"2d36bf6310c7e8eb40f8c84ef23e760463db3c75","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-27 08:45:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/17/810917/2/check/dsvm-tempest-TOYOU-ACS5000-driver/9df7582 : SUCCESS in 2h 42m 21s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/17/810917/2/check/fc-test/cf4219b : SUCCESS in 2h 29m 19s","accounts_in_message":[],"_revision_number":2},{"id":"6f2ff9022bc23935b738aad0c4d4a3ae17c96571","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-12-27 09:00:21.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/810917/2/20211227150139/NEC-FC : SUCCESS in 2h 45m 23s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/810917/2/20211227150139/NEC-ISCSI : SUCCESS in 2h 58m 38s","accounts_in_message":[],"_revision_number":2},{"id":"aae3a5f772d464d0feb02b879fdcb7a448531b20","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-12-27 09:05:40.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/17/810917/2/upstream-solidfire-check/cinder-solidfire-iSCSI/bd3cd98/ : FAILURE in 27m 22s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"05b0dfb12a3acbc1d74f75a7ef2f30fd2017db72","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-27 10:00:47.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-810917-2/ns5_nfs : SUCCESS in 3h 58m 30s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-810917-2/ns5_iscsi : SUCCESS in 3h 57m 34s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"4d409dab62c724573ca55c6e1c0e9ab39e3aca7f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-27 10:17:11.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/17/810917/2/check/DellEMC_XtremIO_FC/d0fe151/DellEMC_XtremIO_FC/361 : SUCCESS in 1h 04m 42s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/2/check/DellEMC_XtremIO_iSCSI/6468f6e/DellEMC_XtremIO_iSCSI/374 : SUCCESS in 1h 39m 52s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"b25dab8f67407b4bfc6da26c5ff671e081cc3896","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-12-27 10:21:34.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/17/810917/2/upstream-check/cinder-cDOT-iSCSI/cb299ed/ : SUCCESS in 1h 43m 04s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/17/810917/2/upstream-check/cinder-cDOT-NFS/a81004c/ : SUCCESS in 1h 43m 34s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/17/810917/2/upstream-check/cinder-cDOT-FCP/df987c3/ : FAILURE in 1h 14m 24s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"b9da55c26a1fd65176ce155b0f7702488b0a6d32","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-12-27 10:34:01.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/17/810917/2/check/DellEMC_SC_FC/a3ade96/DellEMC_SC_FC/143 : SUCCESS in 45m 55s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/2/check/DellEMC_SC_iSCSI/e0f15b5/DellEMC_SC_iSCSI/141 : FAILURE in 1h 56m 43s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"a2319a01ce09818a36c24a15cb5c197afe5033a0","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-12-27 11:05:07.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI http://os-logs.tristero.net/dell-ci/810917-2/PVME-iSCSI : SUCCESS in 58:26.94\n- PVME-FC http://os-logs.tristero.net/dell-ci/810917-2/PVME-FC : SUCCESS in 59:14.03","accounts_in_message":[],"_revision_number":2},{"id":"677987970814a2b19e6548e6a4a7e56a465d2450","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-12-27 11:17:25.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/2/check/DellEMC_VNX_iSCSI/c0fe573/DellEMC_VNX_iSCSI/168 : SUCCESS in 2h 40m 05s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/17/810917/2/check/DellEMC_VNX_FC/f8235d7/DellEMC_VNX_FC/175 : FAILURE in 14m 39s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"ed27645ec41840df08e62a1166d399147a816fb2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-27 14:33:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/7ffd4fbac9c94b669f5867ef79640a9a : RETRY_LIMIT in 2h 15m 39s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/35bef246a7b248f292d32da1cdddc5db : RETRY_LIMIT in 58m 57s","accounts_in_message":[],"_revision_number":2},{"id":"03fec6d1db6c1d137322e8cb6f9d615a9a3f1aba","author":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"date":"2021-12-28 05:01:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bee35175d2919fd0c9c892595bce112cea07308f","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-12-28 07:05:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/82d8d59/index.html : TIMED_OUT in 2h 02m 38s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/4aea29a/index.html : TIMED_OUT in 2h 02m 46s","accounts_in_message":[],"_revision_number":2},{"id":"278674523e400e8c2eed34c7b13fa79ed5faa6df","author":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"date":"2021-12-28 10:42:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ea0ad18d79f1716c6d13db17a4be70399488afd7","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-12-28 12:45:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/a9c07c6/index.html : TIMED_OUT in 2h 02m 42s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/794c150/index.html : TIMED_OUT in 2h 02m 36s","accounts_in_message":[],"_revision_number":2},{"id":"b51b4f84cbdd2305dce6b81dfd7f58634f8154d2","author":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"date":"2021-12-29 15:04:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7855c64232e88ccabb4e6e8cd4bca77c58175f04","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-12-29 17:07:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/edebd1e/index.html : TIMED_OUT in 2h 02m 34s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/f1f4fe4/index.html : TIMED_OUT in 2h 02m 38s","accounts_in_message":[],"_revision_number":2},{"id":"5737dbafe3c2c9035e67012c512098f233f7f785","author":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"date":"2021-12-31 09:53:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e770f990fae0548376135169febaaadbab9128ad","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-12-31 10:59:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/770d1c3/index.html : SUCCESS in 1h 04m 59s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/5ffba7a/index.html : SUCCESS in 1h 04m 25s","accounts_in_message":[],"_revision_number":2},{"id":"6671c431dd2efb12c4f24cbbf00addbb80bec2f4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-04 03:26:39.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/810917/810917/2/check-not-vote/ext-cinder-zuul/49e7e77 : FAILURE in 20m 54s","accounts_in_message":[],"_revision_number":2},{"id":"721189fca13578e222a7d50440949274c618d4e1","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-01-04 21:21:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8755b68ddc4c2ea981171cc2ab4493ba4a171d81","author":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"date":"2022-01-05 07:35:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a552cca348480dcfa420663a406af9e29f6a0c9d","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-01-06 16:49:10.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"54f22c8d8aed747eceb97d7aaefa49a4a5635bcf","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-01-06 17:06:39.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8bf86b7532dd9e9947a9b9647bcadf65a2885ca7","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-01-10 08:15:25.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5eadc796c044e88f1f56c003ce1b4b3dd9226776","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-01-15 06:50:30.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/17/810917/2/check/DellEMC_PowerMAX_FC/d6131dd/DellEMC_PowerMAX_FC/138 : FAILURE in 1h 24m 14s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/2/check/DellEMC_PowerMAX_iSCSI/eaac46d/DellEMC_PowerMAX_iSCSI/139 : FAILURE in 3h 20m 10s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"21036c199d349e0e282248fbf8c09db8d20fe7dd","author":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"date":"2022-01-20 20:02:37.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"05a68ea03c5585ec6a3ee44ed62dc4bf954b7680","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-01-21 15:08:59.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"eb44240079b19b064da7a19e38059ecf73156a3e","author":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"date":"2022-01-24 11:52:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7c067c3291121fcc3f0e6aaec499a603a8b6a209","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-01-28 13:39:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"468e01cec71e0ee35555dcb7fec48331665da816","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-29 07:01:26.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-810917-2/ns5_nfs : SUCCESS in 4h 46m 39s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-810917-2/ns5_iscsi : SUCCESS in 4h 54m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"3a160da9a8082ec13ce4711853cee0fc662b9e82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2022-02-07 13:05:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c43edf2b8a8755324bf0ed1d7a6da810a3365ed5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-07 13:05:49.000000000","message":"Patch Set 3:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15152/","accounts_in_message":[],"_revision_number":3},{"id":"14ee84e8121acbe9a0329d715153e63b9599814c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-07 13:07:35.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15152/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"9a0eddd8a559fdfa4958f1dbff773f9887c816e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 13:29:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2cf000f28d1b46648669d3fc9e6e3e6c : SUCCESS in 17m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e88f2ab0d1b743eeb1f72d003b65ab0f : SUCCESS in 20m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4c0d2868ed33537f2fa49fa0ebd311f57fa7fe61","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-02-07 13:30:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/17/810917/3/check/dsvm-tempest-zadara-driver/ae3c5cd : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 13m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"beb2b3d8b61062248f93f57311fadfddeeb99e0c","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-07 13:41:09.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/17/810917/3/check/DellEMC_XtremIO_FC/10702ab/DellEMC_XtremIO_FC/667 : FAILURE in 14m 38s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/3/check/DellEMC_XtremIO_iSCSI/b243a9c/DellEMC_XtremIO_iSCSI/677 : FAILURE in 15m 54s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"1fdc3ee7621d682b40436209e15a837f29ec318f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-02-07 14:01:31.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/17/810917/3/check/iscsi-driver-primera/ee934a4/ee934a4.html : FAILURE in 51m 21s","accounts_in_message":[],"_revision_number":3},{"id":"e937f97bcbc1e6f089d790bdf1aaa24d200580d7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-07 14:02:50.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/810917/3/iscsi/ : SUCCESS in 57m 03s\n- cinder-smb http://cloudbase-ci.com/cinder/810917/3/smb/ : SUCCESS in 55m 39s","accounts_in_message":[],"_revision_number":3},{"id":"c65927533c31c06fd675400008a044f3cae8ed87","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-07 14:09:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/17/810917/3/check/iscsi-driver-alletra-6k/1f876aa/1f876aa.html : FAILURE in 1h 02m 09s","accounts_in_message":[],"_revision_number":3},{"id":"9db821234f3bc3d9b5fd03532670273956a130e3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-07 14:18:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/17/810917/3/check/dsvm-tempest-cisco-zonemanager-job/211678b/ : SUCCESS in 1h 05m 15s","accounts_in_message":[],"_revision_number":3},{"id":"f87e613991ea19db5ee0606d2df94d964056700d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-07 15:36:29.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/17/810917/3/check/iscsi-driver-nimble/5e2e0b5/5e2e0b5.html : FAILURE in 1h 11m 53s","accounts_in_message":[],"_revision_number":3},{"id":"42707788713f1f1b94b20fe924550243a9971f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 15:50:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec14fc78611b454cb44deb5b0a568acf : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf515cfe399e40b1a09d94aa9ed20fd0 : SUCCESS in 9m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b918ae379f6c4b1d90e8e37977f1a8c9 : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b509761287eb4978aebd80fabb958c70 : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/5f3bfb41a52b44d497ccb128c5ae088b : SUCCESS in 45m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/785394a8f09f4f92953766ca4f975fd3 : SUCCESS in 1h 09m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3752d8bba33941f38c54c3dd9c14143e : SUCCESS in 37m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07758742509a449abeffc2f06bec6f20 : SUCCESS in 7m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/096ef58795854380aba67a50dfd33093 : SUCCESS in 13m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aa69f7839bfd4d24a124e7dd4fb11ab7 : SUCCESS in 5m 14s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5dcd9333408e4f63a12035bdd4a80fca : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c3ec22ddde6e42e09a990511b5e3ee1b : SUCCESS in 7m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dc7dd8077900461d95dd1b242dc896ee : SUCCESS in 7m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9a27765bec824daab22eb0078f5b313d : SUCCESS in 31m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b23df9163fca40b1978dcc8f3a0e8d7d : FAILURE in 7m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/47e8058bc0a945bfa8d3b74501a2bd9e : SUCCESS in 1h 05m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ee4bfb7bcc274d68921d1865c0d97966 : SUCCESS in 1h 23m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/35bbcfbcb5bb4dc7b0fbdadc46946b26 : SUCCESS in 1h 30m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d707ceb2b0cf447ba7dcadc3bb5053bf : SUCCESS in 1h 02m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d91bc2f7cd284d2181800652061f5f24 : SUCCESS in 1h 11m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/73f868e8010b48c59421a2aa74a9e6b3 : SUCCESS in 46m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7167198b8eeb442b9805aa58cfad9420 : SUCCESS in 1h 14m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/01f21cfb139e4796825da9a5f623d1ec : SUCCESS in 2h 32m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d659d4cd02a24bb2946806ff7c85cb78 : SUCCESS in 38m 31s","accounts_in_message":[],"_revision_number":3},{"id":"de5a809c4307807810562e4e5e03434b0ee20ee7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-07 16:05:02.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/17/810917/3/check/DellEMC_PowerFlex/e3c7d24/DellEMC_PowerFlex/374 : FAILURE in 11m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"17e70597cbf36caf2c42ae569e85cd089f1fd726","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2022-02-07 16:05:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eccb405b14515e9099dd2fab9da54eff61dab9e3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-02-07 16:05:09.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/3/check/DellEMC_PowerStore_iSCSI/7fb49b0/DellEMC_PowerStore_iSCSI/324 : FAILURE in 13m 21s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/17/810917/3/check/DellEMC_PowerStore_FC/dbd49b6/DellEMC_PowerStore_FC/341 : FAILURE in 14m 05s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"4328fd01912ef55ec30803e4f23bebb67b1e1555","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-02-07 16:08:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/17/810917/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b60da86/ : SUCCESS in 2h 20m 34s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/17/810917/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/88ce395/ : FAILURE in 3h 01m 21s","accounts_in_message":[],"_revision_number":3},{"id":"adf874e73ef180338006340a11a93b9965832f62","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-07 16:09:05.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/17/810917/3/check/DellEMC_XtremIO_FC/74fbeb9/DellEMC_XtremIO_FC/670 : FAILURE in 14m 51s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/3/check/DellEMC_XtremIO_iSCSI/e0c522f/DellEMC_XtremIO_iSCSI/680 : FAILURE in 15m 51s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"a118bda40fee9c8518029b94d2d23ebaed293528","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-07 16:35:06.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/17/810917/3/check/tempest-dsvm-full-focal-py3/1858312/ : SUCCESS in 1h 52m 59s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/17/810917/3/check/devstack-plugin-ceph-dsvm-focal-py3/c21ed90/ : TIMED_OUT in 3h 14m 05s","accounts_in_message":[],"_revision_number":3},{"id":"ac184b2332ea58af15a955d02b713a4287e05287","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-07 17:05:42.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/17/810917/3/ : SUCCESS in 1h 35m 49s","accounts_in_message":[],"_revision_number":3},{"id":"7a03253f3619e337a89f1adf1e8a734243ef4f70","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-07 17:06:08.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/810917/3/upstream-check/cinder-cDOT-iSCSI/fc42c81/ : NOT_REGISTERED\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/810917/3/upstream-check/cinder-cDOT-NFS/19444ce/ : FAILURE in 1h 12m 50s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/810917/3/upstream-check/cinder-cDOT-FCP/5a444fb/ : FAILURE in 12m 27s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"ad50d5ec6974a1d9ad6a71abfc4cd43797b1088b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-02-07 17:06:47.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/17/810917/3/check/DellEMC_PowerMAX_FC/90d6c6f/DellEMC_PowerMAX_FC/286 : FAILURE in 51m 22s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/3/check/DellEMC_PowerMAX_iSCSI/347f000/DellEMC_PowerMAX_iSCSI/299 : FAILURE in 28m 58s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"a4fab2781fe972c56339b0e6e777abde4ae6fdf2","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-02-07 17:11:06.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/810917/3/upstream-solidfire-check/cinder-solidfire-iSCSI/dea549b/ : FAILURE in 1h 17m 23s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"81b33417e57a164cce9a60043f32789077f22792","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-02-07 17:18:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/8cbdbef/index.html : SUCCESS in 1h 08m 08s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/253fbfa/index.html : SUCCESS in 1h 11m 47s","accounts_in_message":[],"_revision_number":3},{"id":"9a317d77fe287b0807e50ba6461f8fe464a07c86","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2022-02-07 17:55:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1927423217bfd7fa97a91039e74cacd6be58a1fb","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-02-07 19:10:18.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/17/810917/3/check/DellEMC_SC_FC/dbc8c6f/DellEMC_SC_FC/308 : FAILURE in 1h 08m 09s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/3/check/DellEMC_SC_iSCSI/db4afaa/DellEMC_SC_iSCSI/304 : SUCCESS in 1h 47m 25s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"8302057f081c0c5e8dc9f3fb86412d60c847240c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-02-07 19:55:27.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/810917/3/20220208040316/NEC-V-ISCSI : FAILURE in 48m 29s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/810917/3/20220208040316/NEC-V-FC : FAILURE in 52m 08s","accounts_in_message":[],"_revision_number":3},{"id":"337bffc1051f4be756d7ed31a6a645d4cf9e21d0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-07 20:13:17.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-810917-3/ns5_nfs : SUCCESS in 4h 41m 17s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-810917-3/ns5_iscsi : FAILURE in 53m 33s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"843bcf5ed73fd40fd099b116a8bdbcab33f76df6","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-02-08 15:46:04.000000000","message":"Patch Set 3: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"689215af1e11c6742b4a1a844b5ca6114a193b5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2022-02-09 11:55:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3c689a2b8e7fda522ad938f5ec69715ba4643efb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-09 11:55:23.000000000","message":"Patch Set 4:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15167/","accounts_in_message":[],"_revision_number":4},{"id":"ea7d2476aaaf2d552399fb6339ed45a30c2ca086","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-09 11:57:49.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15167/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"7202ec2cca6776d5f5910bac8e26e8a08416fb6a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-09 12:10:51.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/810917/810917/4/check-not-vote/ext-cinder-zuul/4d68418 : FAILURE in 6m 23s","accounts_in_message":[],"_revision_number":4},{"id":"5905a952a97f8d117094a9debada9705f5027137","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-09 12:13:17.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/17/810917/4/check/DellEMC_XtremIO_FC/afd8e8d/DellEMC_XtremIO_FC/698 : FAILURE in 14m 50s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/4/check/DellEMC_XtremIO_iSCSI/7aca6f8/DellEMC_XtremIO_iSCSI/708 : FAILURE in 14m 58s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"37aa15c362b57fc23e83a716a95bb420bcd0724f","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2022-02-09 12:16:09.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9797f3c7fd710464a1cd54abc09a30e809c20aa1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-02-09 12:21:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/810917/3/20220209175641/NEC-FC : SUCCESS in 3h 22m 55s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/810917/3/20220209175641/NEC-ISCSI : SUCCESS in 3h 23m 32s","accounts_in_message":[],"_revision_number":3},{"id":"d615e24c00eee72e8d3f9b55633ca4dc28b1edf3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 12:34:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/66fd802e827c4aa0b46f43fcf5f57a1a : SUCCESS in 19m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57f02266f1894d759d04c3e12ab97d47 : SUCCESS in 23m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c98e6c8e301b4e501374a8eff3c4a96b5c48dd12","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-02-09 13:02:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/17/810917/4/check/dsvm-tempest-zadara-driver/b6af2a3 : SUCCESS in 1h 04m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5f45a1887f1c6a7a2440ebf9fd103884f2d995fb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-09 13:06:34.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/17/810917/4/check/dsvm-tempest-cisco-zonemanager-job/f9c98ff/ : SUCCESS in 1h 04m 54s","accounts_in_message":[],"_revision_number":4},{"id":"a6c3131ff239f9c33c039e5f804bf6226a1f5d3b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-09 13:07:06.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/810917/4/iscsi/ : SUCCESS in 1h 11m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/810917/4/smb/ : SUCCESS in 1h 04m 39s","accounts_in_message":[],"_revision_number":4},{"id":"2dca864dc8cda9f3cc26472b857fa8a4c66863ef","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-02-09 13:29:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/17/810917/4/check/iscsi-driver-primera/d1b3825/d1b3825.html : SUCCESS in 1h 29m 00s","accounts_in_message":[],"_revision_number":4},{"id":"2a8c3004284b3e9301f21e7a2963abbf2ca94256","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-09 13:35:32.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/17/810917/4/check/iscsi-driver-alletra-6k/9b9510c/9b9510c.html : SUCCESS in 1h 39m 13s","accounts_in_message":[],"_revision_number":4},{"id":"cbdecf418a2cdd8eb29a61d55e6bd9027ad564b7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-09 13:52:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/17/810917/4/check/iscsi-driver-nimble/1ae5e07/1ae5e07.html : SUCCESS in 1h 55m 41s","accounts_in_message":[],"_revision_number":4},{"id":"0d1b58a7815e462521682b416023dd73a1c8055b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-02-09 14:35:09.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/17/810917/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/71f3c1f/ : SUCCESS in 1h 39m 48s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/17/810917/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a78549f/ : FAILURE in 2h 38m 07s","accounts_in_message":[],"_revision_number":4},{"id":"bd61133a651b3ad3ad88bc7624b5ff2c7b8b4596","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 14:35:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af1591666cc54fdc8d03e09dd0248a6f : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb046fb1a3ba426f88977d4a8f18b11d : SUCCESS in 11m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fce38366170a411eb62037d962c92096 : SUCCESS in 12m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07c9972c43fb4471b02c8f5ea59fb19a : SUCCESS in 13m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/a849b1ba8fdb45a98a4ec4737fe56c5a : SUCCESS in 57m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/54d011c3d57444e08b4daff239ee1287 : SUCCESS in 1h 06m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7f6ce5bf5fe4d369b175c4b5f166db3 : SUCCESS in 48m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a93a624cc5df4aa68bcf23054d7f1bca : SUCCESS in 8m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9ad9e6602ef14b9495d6830d5c77eb90 : SUCCESS in 15m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3d279d72a5ae43d88dd48dd63d9d49f1 : SUCCESS in 7m 27s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a011af9780d143588b20aac943322e8c : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ed9ec15cacd241b5806b4750970c84a0 : SUCCESS in 7m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1eab461bb0a9416d83deda902cc4e2f0 : SUCCESS in 9m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/76fb85b915e24d21be6d77b67536981e : SUCCESS in 39m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eb006fea9f3e405383157d5da2695738 : FAILURE in 7m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1ede133af01149748801a00c323eea83 : SUCCESS in 1h 20m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/97797b3d24934a8f9b9614c97837a454 : SUCCESS in 1h 21m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ef8d2475fab54d55aba8f0963a7a6379 : SUCCESS in 1h 46m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/03233e94e2ee4ae2a38b36345d843ab0 : SUCCESS in 59m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c1559cdd680341dba9b6aa1c8aa30926 : FAILURE in 1h 20m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/df99236dd27d4c3a82d1dc836eebdfb8 : SUCCESS in 50m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d4e94161429e4f518f290f45c768da50 : SUCCESS in 1h 16m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bd18cb345c4e48eeadebd3db7465c636 : SUCCESS in 2h 20m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec9ffab9898848f69465e1929fea924b : SUCCESS in 47m 29s","accounts_in_message":[],"_revision_number":4},{"id":"7b50228f95778cb001982a243d7bed419614f547","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2022-02-09 14:38:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ca8c5ae936482361e9444d8eed4a97c28dad4990","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-02-09 14:50:36.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/4/check/DellEMC_PowerStore_iSCSI/cd3dfb8/DellEMC_PowerStore_iSCSI/338 : FAILURE in 13m 41s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/17/810917/4/check/DellEMC_PowerStore_FC/bf29382/DellEMC_PowerStore_FC/355 : FAILURE in 14m 32s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"8a96e722ef9096c1ad0f808f256b8e1e48d1fd72","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-09 14:52:13.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/17/810917/4/check/tempest-dsvm-full-focal-py3/45735af/ : SUCCESS in 1h 55m 22s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/17/810917/4/check/devstack-plugin-ceph-dsvm-focal-py3/e3d9cbb/ : SUCCESS in 2h 36m 21s","accounts_in_message":[],"_revision_number":4},{"id":"08d1255c04d5eab673fb268b278f7748b35290ef","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-09 14:54:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-17-810917-4/1644417345 : SUCCESS in 18m 39s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"df675f020e9b281c55d5111816dadedf34d5a425","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-09 14:57:57.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/17/810917/4/check/DellEMC_PowerFlex/2f74ca4/DellEMC_PowerFlex/389 : FAILURE in 12m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"58a49d63c783bcbb784b7cc2d9f637954f83b943","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-09 15:01:08.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/17/810917/4/check/DellEMC_XtremIO_FC/947a387/DellEMC_XtremIO_FC/700 : FAILURE in 14m 46s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/4/check/DellEMC_XtremIO_iSCSI/9f4ce40/DellEMC_XtremIO_iSCSI/710 : FAILURE in 15m 03s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"8b9fff0fc55f8102689bfdb09f8df07f16271c99","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-09 15:13:30.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/17/810917/4/ : SUCCESS in 1h 41m 37s","accounts_in_message":[],"_revision_number":4},{"id":"9d11ceaeb5e0f91fbcd34dfd821a281689f9e44e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-02-09 15:38:13.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/17/810917/4/check/DellEMC_PowerMAX_FC/186aa4c/DellEMC_PowerMAX_FC/300 : FAILURE in 52m 28s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/4/check/DellEMC_PowerMAX_iSCSI/6a14ebb/DellEMC_PowerMAX_iSCSI/313 : FAILURE in 27m 01s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"037908caa9d1a78f8c6da3522fce9e37fd0616b2","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-02-09 15:51:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/260cef3/index.html : SUCCESS in 1h 06m 15s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/fcffd89/index.html : SUCCESS in 1h 12m 43s","accounts_in_message":[],"_revision_number":4},{"id":"fb2a1339a2c830fda3b0d037011b3d49da306f61","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-09 16:41:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/810917-4/HPMSA-iSCSI/ : SUCCESS in 50:31.54\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/810917-4/HPMSA-FC/ : SUCCESS in 43:42.34","accounts_in_message":[],"_revision_number":4},{"id":"2749c1b3754b4f4ed549fd10dc42f72aabffd149","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-02-09 17:13:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/810917-4/Seagate-iSCSI/ : SUCCESS in 1:03:12\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/810917-4/Seagate-FC/ : SUCCESS in 1:02:59","accounts_in_message":[],"_revision_number":4},{"id":"2356dc23cfc214a3f7720bcb20adcdb4c600e44e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-09 17:25:17.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-810917-4/ns5_nfs : SUCCESS in 4h 45m 51s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-810917-4/ns5_iscsi : SUCCESS in 4h 49m 33s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"73e25e2b50d92dbf533936e9d1655d8ac23f9616","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-09 17:55:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/810917-4/Lenovo-iSCSI/ : SUCCESS in 1:03:05\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/810917-4/Lenovo-FC/ : SUCCESS in 57:17.52","accounts_in_message":[],"_revision_number":4},{"id":"b8e2485e5384705cc51a602a9e033e45e7a53260","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-02-09 18:11:29.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/17/810917/4/check/DellEMC_SC_FC/d5163d9/DellEMC_SC_FC/322 : FAILURE in 1h 10m 07s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/4/check/DellEMC_SC_iSCSI/afaafcf/DellEMC_SC_iSCSI/318 : SUCCESS in 1h 45m 35s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"f3251f7e71f0885c8146450980f3cb9ba7ee689d","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-09 20:22:50.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/810917-4/PVME-iSCSI/ : SUCCESS in 1:01:45\n- PVME-FC https://os-logs.seagate.com/dell-ci/810917-4/PVME-FC/ : SUCCESS in 1:00:18","accounts_in_message":[],"_revision_number":4},{"id":"0e98a332f3c878e43ee91157b5dbd0bf2324b4da","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-02-09 21:30:31.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/810917/4/20220210053847/NEC-V-ISCSI : FAILURE in 48m 16s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/810917/4/20220210053847/NEC-V-FC : FAILURE in 51m 40s","accounts_in_message":[],"_revision_number":4},{"id":"49de34578c45ebacd2f23076db75fb34b90b57a4","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-02-10 14:57:37.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a117575a6b51979aced7abc336a0c4e5cba79a32","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-02-11 06:47:52.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/810917/4/20220211122738/NEC-ISCSI : SUCCESS in 3h 19m 45s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/810917/4/20220211122738/NEC-FC : SUCCESS in 3h 20m 11s","accounts_in_message":[],"_revision_number":4},{"id":"6b9e07140ad71d10abdc1953e72e302654e25376","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2022-02-11 12:27:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1dda4adbd7d06f745b7a9307be27a0ee4ecef9be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 12:59:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1782c1e24d947ff80b6469a26849121 : FAILURE in 21m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b12fc0ff0db47e5bf1b055c3c770c40 : FAILURE in 19m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3938c1e56f2ede6b8b918bd85393ae8e9143cafa","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-11 13:23:09.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/810917/5/iscsi/ : SUCCESS in 55m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/810917/5/smb/ : SUCCESS in 52m 34s","accounts_in_message":[],"_revision_number":5},{"id":"204eaaec8c4fc585364ece2939026857514ebaa9","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-02-11 13:47:02.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/17/810917/5/check/dsvm-tempest-zadara-driver/15f49e0 : SUCCESS in 1h 12m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ccad678563d16a26a0d2088d63f5dbbb4618d355","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-02-11 13:49:36.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/17/810917/5/check/iscsi-driver-primera/4272062/4272062.html : SUCCESS in 1h 20m 31s","accounts_in_message":[],"_revision_number":5},{"id":"8023a1a928e160088080666eff660e8024ed56db","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-11 14:05:27.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/17/810917/5/check/iscsi-driver-alletra-6k/e54e390/e54e390.html : SUCCESS in 1h 36m 58s","accounts_in_message":[],"_revision_number":5},{"id":"32b11c210dea1cc56e95ab52bd2506e1c29b9c90","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-02-11 14:40:13.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/17/810917/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/7a924e8/ : SUCCESS in 2h 09m 07s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/17/810917/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b14f7cf/ : SUCCESS in 2h 09m 52s","accounts_in_message":[],"_revision_number":5},{"id":"2ae8dface2a4fb32d08e05e262b6b62f27cff8ae","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2022-02-11 14:47:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"25ffa6278e707a5731a95b378ff9144cc2f62bec","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-11 14:49:42.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/17/810917/5/check/DellEMC_XtremIO_FC/f99d8bc/DellEMC_XtremIO_FC/732 : SUCCESS in 1h 00m 38s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/5/check/DellEMC_XtremIO_iSCSI/070960f/DellEMC_XtremIO_iSCSI/740 : FAILURE in 16m 27s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"27ea61378655039968e5522e850729ac8e044798","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-11 14:52:05.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/17/810917/5/check/dsvm-tempest-cisco-zonemanager-job/993d90d/ : SUCCESS in 1h 10m 54s","accounts_in_message":[],"_revision_number":5},{"id":"1a262cb47b0ae7ecc64426f56909c921ec82c0f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 15:19:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d742b2d6cb542ebbdaf5df1b8539363 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49c562d4f3f84aed81cecd223e49e131 : SUCCESS in 9m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/677f74885c3246eab5b76b8c097ef255 : SUCCESS in 10m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3dd72aa3d244f64aca4907f98471698 : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/1608897d8739460d861ee4f6eddd42d9 : SUCCESS in 53m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1e0697fd022d4a36ac9d5f067f423ea2 : SUCCESS in 1h 01m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/601e16f6ba7949959632487a25fe47d2 : SUCCESS in 56m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d58ee852baee4afaa27f7966cacc939f : SUCCESS in 7m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d24d926a458445a99497ffe870aca64b : SUCCESS in 14m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/633113cc88ad4f028c5ad73258401767 : SUCCESS in 6m 46s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/df2ac8877f20432d91b637e6f98f7fdf : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a22c394027bf4a739be35a95c98d64e2 : SUCCESS in 6m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ade3bbafeda24c649032f65118a6c342 : SUCCESS in 7m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a8cdf1a80f184db1a628ccb7978d6d19 : SUCCESS in 53m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2322185fe04745ed866cab6b303783b1 : FAILURE in 7m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a0c5fb389a1442bf9cfe41dc40b29671 : SUCCESS in 1h 12m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6e7633b69c434245a6485b1c73a0aaad : SUCCESS in 1h 27m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2e257753e7af4d92971ec65435997c07 : SUCCESS in 1h 50m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3a6493423e034795b0f5069e1d322f0c : SUCCESS in 1h 06m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7e5a9c681d724e52b1004de0924d4c18 : SUCCESS in 1h 27m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a56d4a02bfd34eafaff4f0b5f37007f0 : SUCCESS in 44m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/562281a71c324bd28d6ff0ab2c8714fd : FAILURE in 1h 12m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/52bf82d3b0cd4deb9cbb9392e8c20dce : SUCCESS in 2h 44m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59b34f79a93d4bdf9a0d7f259a2fac08 : SUCCESS in 49m 22s","accounts_in_message":[],"_revision_number":5},{"id":"94e5682fc0092cb575b30e56976c2ca6c162eca2","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-02-11 15:21:13.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/810917/5/upstream-solidfire-check/cinder-solidfire-iSCSI/9aa2596/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"96676abf3294806cc0efbb01cb66cfe6b4695b12","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-11 15:24:26.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/17/810917/5/check/tempest-dsvm-full-focal-py3/1d6ff72/ : SUCCESS in 1h 53m 56s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/17/810917/5/check/devstack-plugin-ceph-dsvm-focal-py3/f7cfc94/ : SUCCESS in 2h 28m 06s","accounts_in_message":[],"_revision_number":5},{"id":"cc1e5b954d3d2034cec3f2db15a25024606bbf25","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-11 15:30:41.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/17/810917/5/ : SUCCESS in 1h 58m 11s","accounts_in_message":[],"_revision_number":5},{"id":"f0a9fb838c443c588d80af585607ec0b47c0c485","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-02-11 15:31:30.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/810917/5/20220211233758/NEC-V-ISCSI : FAILURE in 51m 29s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/810917/5/20220211233758/NEC-V-FC : FAILURE in 53m 29s","accounts_in_message":[],"_revision_number":5},{"id":"94cf733b56e385e865d2b0e8ff669db344242fc3","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-11 15:37:09.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-17-810917-5/1644592771 : SUCCESS in 17m 36s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":5},{"id":"97d598b6b83fc4c8cbfedd432993eb95aef232f0","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2022-02-11 15:37:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b3e887fee8ac1f20e42dda6c2c4c87e33f0544d4","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-02-11 16:11:18.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/17/810917/5/check/DellEMC_PowerMAX_FC/8b56884/DellEMC_PowerMAX_FC/314 : FAILURE in 51m 28s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/5/check/DellEMC_PowerMAX_iSCSI/e45e617/DellEMC_PowerMAX_iSCSI/327 : FAILURE in 26m 07s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"f85a43d8db0cd0763923a80a6c12a5bcb820bbf1","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-11 16:41:53.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/17/810917/5/check/DellEMC_XtremIO_FC/822b8da/DellEMC_XtremIO_FC/733 : SUCCESS in 59m 46s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/17/810917/5/check/DellEMC_XtremIO_iSCSI/54e3aa0/DellEMC_XtremIO_iSCSI/741 : FAILURE in 15m 27s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"62b180bd530a5175de8d192415abe109d522a96e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-11 16:43:22.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-810917-5/ns5_nfs : SUCCESS in 4h 7m 4s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-810917-5/ns5_iscsi : SUCCESS in 4h 15m 24s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"01e952a0a089e370dbbb886199a2d680be71f927","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-02-11 16:54:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/6bcd5e4/index.html : SUCCESS in 1h 09m 35s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/69ab51d/index.html : SUCCESS in 1h 15m 44s","accounts_in_message":[],"_revision_number":5},{"id":"2817bd15677d62a441b59d1247a3dd5bb79de943","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-11 17:05:59.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/810917/5/upstream-check/cinder-cDOT-iSCSI/09aa726/ : SUCCESS in 1h 43m 00s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/810917/5/upstream-check/cinder-cDOT-NFS/900be89/ : SUCCESS in 1h 45m 08s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/810917/5/upstream-check/cinder-cDOT-FCP/2f665b3/ : SUCCESS in 51m 37s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"866e67951462c88ee6db21ffdebcfac318971761","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-11 17:08:30.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/17/810917/5/check/DellEMC_PowerFlex/be58035/DellEMC_PowerFlex/408 : FAILURE in 1h 48m 37s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"3a4b4b48cda1fa47d3ee32fa38a79f76853f680d","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-11 17:24:13.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/810917-5/HPMSA-iSCSI/ : FAILURE in 47:08.89\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/810917-5/HPMSA-FC/ : SUCCESS in 42:58.44","accounts_in_message":[],"_revision_number":5},{"id":"a726d4ed68900289bd11afb9cc9ca596501592a9","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-11 17:53:00.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/810917-5/PVME-iSCSI/ : SUCCESS in 1:01:34\n- PVME-FC https://os-logs.seagate.com/dell-ci/810917-5/PVME-FC/ : SUCCESS in 1:01:43","accounts_in_message":[],"_revision_number":5},{"id":"384662958ec1a105e915c3cf97d77229409323cc","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-11 17:53:15.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/810917-5/Lenovo-iSCSI/ : SUCCESS in 1:04:48\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/810917-5/Lenovo-FC/ : SUCCESS in 56:40.07","accounts_in_message":[],"_revision_number":5},{"id":"ff17996a509b5b5dbe75c6c56be30c5572a26ed8","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-02-11 18:01:16.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/810917-5/Seagate-iSCSI/ : SUCCESS in 1:02:01\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/810917-5/Seagate-FC/ : SUCCESS in 1:03:59","accounts_in_message":[],"_revision_number":5},{"id":"84233e43d943cd77d4b890fa9a216d7ac690ef6c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-02-13 07:52:17.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/810917/5/20220213132838/NEC-ISCSI : SUCCESS in 3h 13m 38s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/810917/5/20220213132838/NEC-FC : SUCCESS in 3h 23m 35s","accounts_in_message":[],"_revision_number":5},{"id":"0b13fb37e1c6f656bf41af0a92f764f472705a7e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-14 01:20:54.000000000","message":"Patch Set 5:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":5},{"id":"4853c2516619a201389694e2c171c69e4efbcfa3","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-02-14 12:29:10.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7316bb20cd441700c8a2d2b8fdbde90888da3e56","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-02-14 15:18:24.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9d113774ece94c010299d568a6d1b6b9ddae26e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 15:19:11.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4cb90b6d7fa517d1ef38f579db2780354a5b416e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 16:55:12.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb72d3d4b07b4dc3b7a8143cd9058842 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e03e2bd0d61f4474af9d2b2628975952 : SUCCESS in 10m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a847b756c1094e1fa406a4b4183356fd : SUCCESS in 11m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35fd384378934ef2b69a35f9f28c0ec6 : SUCCESS in 11m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/39faabc7cbea4b43b78dcfa02d54caa7 : SUCCESS in 58m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b4cf8a21fea490dbebc9ca6965e2281 : SUCCESS in 1h 28m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87fb63edec0d4531abf917c8b9af9606 : SUCCESS in 1h 02m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9ef401db04a4020a1c5b0e9cbd87a04 : SUCCESS in 6m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/03807f169c37446ab10fe129dd9dbf3b : SUCCESS in 1h 12m 11s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6839df694e97429d8648d915ce46991b : SUCCESS in 1h 15m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87ae52f06ca346378d7b869f32206b81 : SUCCESS in 56m 24s","accounts_in_message":[],"_revision_number":5},{"id":"874cefb4c595fa9229918bd4e65a470900e11a05","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 16:55:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"f4b1ef7805b8426c3b28182e81459f30e0477964","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 16:57:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe7b9e44b96445efb8c858e5cc18fd03 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a88c1368e869479084ef56437de0d0cb : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"88d3005df43e00ac2d93b70be5bc696b1ad3a275","revisions":{"e9f7006c9189f4bfe667c58291319d08dd59cb6f":{"kind":"REWORK","_number":1,"created":"2021-09-24 14:05:33.000000000","uploader":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"ref":"refs/changes/17/810917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/810917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/810917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/810917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/810917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/810917/1"}}},"commit":{"parents":[{"commit":"a8da1fa83b8330a533a425ec6260bcbb5a3b4f67","subject":"Merge \"Specify xena os-brick series in requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8da1fa83b8330a533a425ec6260bcbb5a3b4f67"}]}],"author":{"name":"Venkata Krishna","email":"venkata.krishna.reddy@ibm.com","date":"2021-09-24 14:00:02.000000000","tz":0},"committer":{"name":"Venkata Krishna","email":"venkata.krishna.reddy@ibm.com","date":"2021-09-24 14:04:45.000000000","tz":0},"subject":"[SVF] update rccg name property to metadata","message":"[SVF] update rccg name property to metadata\n\n[Spectrum Virtualize Family] While creating a group either from\nsource_group or group_snapshot, the resulted group-volumes metadata\nis not updated with the rccg_name attribute. This is happening due to\nthe order of some function calls to update volume metadata.\n\nProviding a fix to update rccg_name property to volume metadata which\ncould cover all the possible cases like add a volume to the group,\ncreate a group from source_group and create a group from group_snapshot.\n\nCloses-Bug: #1943682\n\nChange-Id: I0bfb79db4f69c0705a1e0fa971376d14c232d5cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9f7006c9189f4bfe667c58291319d08dd59cb6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9f7006c9189f4bfe667c58291319d08dd59cb6f"}]},"branch":"refs/heads/master"},"bb9ba48bb0e121cc6a14a10510bc6f45c48e3413":{"kind":"REWORK","_number":2,"created":"2021-12-27 06:01:35.000000000","uploader":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"ref":"refs/changes/17/810917/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/810917/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/810917/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/810917/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/810917/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/810917/2"}}},"commit":{"parents":[{"commit":"de6d54108b4152beee952ff5789861811dbe0016","subject":"Merge \"Use \u0027visibility\u0027 instead of \u0027is_public\u0027 when only use Glance API v2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de6d54108b4152beee952ff5789861811dbe0016"}]}],"author":{"name":"Venkata Krishna","email":"venkata.krishna.reddy@ibm.com","date":"2021-09-24 14:00:02.000000000","tz":0},"committer":{"name":"venkatakrishnathumu","email":"venkata.krishna.reddy@ibm.com","date":"2021-12-27 06:01:00.000000000","tz":0},"subject":"[SVF] update rccg name property to metadata","message":"[SVF] update rccg name property to metadata\n\n[Spectrum Virtualize Family] While creating a group either from\nsource_group or group_snapshot, the resulted group-volumes metadata\nis not updated with the rccg_name attribute. This is happening due to\nthe order of some function calls to update volume metadata.\n\nProviding a fix to update rccg_name property to volume metadata which\ncould cover all the possible cases like add a volume to the group,\ncreate a group from source_group and create a group from group_snapshot.\n\nCloses-Bug: #1943682\n\nChange-Id: I0bfb79db4f69c0705a1e0fa971376d14c232d5cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bb9ba48bb0e121cc6a14a10510bc6f45c48e3413"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bb9ba48bb0e121cc6a14a10510bc6f45c48e3413"}]},"branch":"refs/heads/master"},"fd610c7aa9214388a1c46b3fe8c9e1c710fcfda8":{"kind":"REWORK","_number":3,"created":"2022-02-07 13:05:15.000000000","uploader":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"ref":"refs/changes/17/810917/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/810917/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/810917/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/810917/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/810917/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/810917/3"}}},"commit":{"parents":[{"commit":"de6d54108b4152beee952ff5789861811dbe0016","subject":"Merge \"Use \u0027visibility\u0027 instead of \u0027is_public\u0027 when only use Glance API v2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de6d54108b4152beee952ff5789861811dbe0016"}]}],"author":{"name":"Venkata Krishna","email":"venkata.krishna.reddy@ibm.com","date":"2021-09-24 14:00:02.000000000","tz":0},"committer":{"name":"sreerammounika","email":"sreeram.mounika@ibm.com","date":"2022-02-07 13:01:26.000000000","tz":0},"subject":"[SVf] update rccg name property to metadata","message":"[SVf] update rccg name property to metadata\n\n[Spectrum Virtualize family] While creating a group either from\nsource_group or group_snapshot, the resulted group-volumes metadata\nis not updated with the rccg_name attribute. This is happening due to\nthe order of some function calls to update volume metadata.\n\nProviding a fix to update rccg_name property to volume metadata which\ncould cover all the possible cases like add a volume to the group,\ncreate a group from source_group and create a group from group_snapshot.\n\nCloses-Bug: #1943682\n\nChange-Id: I0bfb79db4f69c0705a1e0fa971376d14c232d5cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fd610c7aa9214388a1c46b3fe8c9e1c710fcfda8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fd610c7aa9214388a1c46b3fe8c9e1c710fcfda8"}]},"branch":"refs/heads/master"},"ece9a904d0f239ba0c32867b85f7dfc9604f3007":{"kind":"REWORK","_number":4,"created":"2022-02-09 11:55:10.000000000","uploader":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"ref":"refs/changes/17/810917/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/810917/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/810917/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/810917/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/810917/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/810917/4"}}},"commit":{"parents":[{"commit":"de6d54108b4152beee952ff5789861811dbe0016","subject":"Merge \"Use \u0027visibility\u0027 instead of \u0027is_public\u0027 when only use Glance API v2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de6d54108b4152beee952ff5789861811dbe0016"}]}],"author":{"name":"Venkata Krishna","email":"venkata.krishna.reddy@ibm.com","date":"2021-09-24 14:00:02.000000000","tz":0},"committer":{"name":"sreerammounika","email":"sreeram.mounika@ibm.com","date":"2022-02-09 11:53:34.000000000","tz":0},"subject":"[SVf] update rccg name property to metadata","message":"[SVf] update rccg name property to metadata\n\n[Spectrum Virtualize family] While creating a group either from\nsource_group or group_snapshot, the resulted group-volumes metadata\nis not updated with the rccg_name attribute. This is happening due to\nthe order of some function calls to update volume metadata.\n\nProviding a fix to update rccg_name property to volume metadata which\ncould cover all the possible cases like add a volume to the group,\ncreate a group from source_group and create a group from group_snapshot.\n\nCloses-Bug: #1943682\n\nChange-Id: I0bfb79db4f69c0705a1e0fa971376d14c232d5cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ece9a904d0f239ba0c32867b85f7dfc9604f3007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ece9a904d0f239ba0c32867b85f7dfc9604f3007"}]},"branch":"refs/heads/master"},"88d3005df43e00ac2d93b70be5bc696b1ad3a275":{"kind":"REWORK","_number":5,"created":"2022-02-11 12:27:25.000000000","uploader":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"ref":"refs/changes/17/810917/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/810917/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/810917/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/810917/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/810917/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/810917/5"}}},"commit":{"parents":[{"commit":"de6d54108b4152beee952ff5789861811dbe0016","subject":"Merge \"Use \u0027visibility\u0027 instead of \u0027is_public\u0027 when only use Glance API v2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de6d54108b4152beee952ff5789861811dbe0016"}]}],"author":{"name":"Venkata Krishna","email":"venkata.krishna.reddy@ibm.com","date":"2021-09-24 14:00:02.000000000","tz":0},"committer":{"name":"sreerammounika","email":"sreeram.mounika@ibm.com","date":"2022-02-11 12:24:39.000000000","tz":0},"subject":"[SVf] update rccg name property to metadata","message":"[SVf] update rccg name property to metadata\n\n[Spectrum Virtualize family] While creating a group either from\nsource_group or group_snapshot, the resulted group-volumes metadata\nis not updated with the rccg_name attribute. This is happening due to\nthe order of some function calls to update volume metadata.\n\nProviding a fix to update rccg_name property to volume metadata which\ncould cover all the possible cases like add a volume to the group,\ncreate a group from source_group and create a group from group_snapshot.\n\nCloses-Bug: #1943682\n\nChange-Id: I0bfb79db4f69c0705a1e0fa971376d14c232d5cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88d3005df43e00ac2d93b70be5bc696b1ad3a275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88d3005df43e00ac2d93b70be5bc696b1ad3a275"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
