)]}'
{"id":"openstack%2Fcinder~811085","triplet_id":"openstack%2Fcinder~master~I1bb83cb271049094e35299a2bdbb54dcb6d8c29c","project":"openstack/cinder","branch":"master","topic":"Bug1945087","hashtags":[],"change_id":"I1bb83cb271049094e35299a2bdbb54dcb6d8c29c","subject":"check the status of backup","status":"NEW","created":"2021-09-27 07:49:12.000000000","updated":"2021-10-03 17:16:22.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a45732404feec2557a2476e9edd1e775688db89f","_number":811085,"virtual_id_number":811085,"owner":{"_account_id":33892,"name":"silphon zhang","email":"zhangxiaofan02@inspur.com","username":"zhangxiaofan02"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-27 09:36:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-09-29 04:47:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":-1,"date":"2021-09-29 15:11:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-27 08:15:48.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-27 08:16:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-27 08:58:01.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-09-27 09:24:44.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-27 09:26:58.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-09-27 09:36:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-27 09:59:42.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-09-27 10:54:11.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-09-27 11:57:14.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-29 04:47:49.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-09-29 15:11:42.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-10-01 12:22:29.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-03 17:16:22.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"cb82aaa8a3182a7537af9512c4f2c3e33e59fb1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33892,"name":"silphon zhang","email":"zhangxiaofan02@inspur.com","username":"zhangxiaofan02"},"date":"2021-09-27 07:49:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78a0f73d2b564a957d5184e03ca4e65f2dace1bd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-27 08:15:48.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/85/811085/1/check/iscsi-driver-alletra-6k/990f491/990f491.html : FAILURE in 25m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b239cbb7147326c7358e7079676c3dc50a6afa9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 08:16:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/36f0361bbab54fbcaf6dac7c9640116c : SUCCESS in 21m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fad72b60dad34864ae191ab65b501c32 : SUCCESS in 19m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a73e93e1203ed5a824703190cafe9ea784bc5b6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-09-27 08:58:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/811085/1/iscsi/ : SUCCESS in 1h 08m 08s\n- cinder-smb http://cloudbase-ci.com/cinder/811085/1/smb/ : SUCCESS in 1h 04m 20s","accounts_in_message":[],"_revision_number":1},{"id":"68dd0c224b4714d2cc32ae5732f2bf926850a6ea","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-27 09:08:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/85/811085/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/cb686f5/cb686f5.html : SUCCESS in 1h 18m 06s","accounts_in_message":[],"_revision_number":1},{"id":"36c57c9b4d762a28091b5d5e38b0fea307fd4a9f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-09-27 09:24:44.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/811085/811085/1/check-not-vote/ext-cinder-zuul/47214f6 : SUCCESS in 31m 24s","accounts_in_message":[],"_revision_number":1},{"id":"884628f4595bc1f82b78c40e83245e007037b405","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-09-27 09:26:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/85/811085/1/check/dsvm-tempest-cisco-zonemanager-job/efb0095/ : SUCCESS in 1h 08m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b1cc636d0574c695197824e197cd0a784a991e88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 09:36:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ed8a5602b154f5aa54305cf138b1b54 : FAILURE in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/517753dbdbdb48d681cddb01afd1f399 : SUCCESS in 15m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e947791bee3d4ab9867bfe0958aa002b : SUCCESS in 11m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31a64c0ccb2743e2b9781b0da79046d8 : SUCCESS in 10m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/534f654b69f84aee947abaf4eb33b726 : SUCCESS in 12m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/60c9061bb3bc442cbedf9de334d0b780 : SUCCESS in 1h 11m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/01b80657cfdb426da51ab176060b0254 : SUCCESS in 1h 06m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/601ff6c1163049f3a4a3a3f6bd617f6a : SUCCESS in 12m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d96a9a59c089456e9243e42db59fd867 : SUCCESS in 7m 31s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0f5fea59591b41fbb4b8fc50f4ee0f12 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7975bb7b2c3a4d7dac5c5e9f1d0a2d95 : SUCCESS in 7m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c32dc6d4180649f2ae2c7306d81ac2d6 : SUCCESS in 9m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a1a96fc8c0584edc8f5035e0191bad93 : SUCCESS in 37m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b23444803c4c4b7c930ccd642b20cc31 : SUCCESS in 6m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a564468d68e749c5a6e32a6ccb66b8f5 : SUCCESS in 1h 15m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2b8f15bfad6640f3abea27cdb6d0d533 : SUCCESS in 1h 24m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d2f74b56692b491a8fc25990e0613b93 : SUCCESS in 1h 29m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/da76d23009cc4f4a838805ac660fc0bd : SUCCESS in 1h 07m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9597ae2aa2614d42b4c631c67623e3c0 : SUCCESS in 1h 20m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1adafc7762bb4182ab028ef28c808ebe : SUCCESS in 42m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8390b8a9710b4196bb79db06c843dcaf : SUCCESS in 1h 09m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b0b185034e9d4202baf381abb8239bd0 : SUCCESS in 1h 41m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b05b775663af462898cc5a81018f38e0 : SUCCESS in 50m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b94ab52e916a4f399ec3f9c3283b460f : SUCCESS in 49m 46s\n\nWarning:\n  Comments left for invalid file cinder/backup/manager.py","accounts_in_message":[],"_revision_number":1},{"id":"d23a6297d2c9311cb47c969c35a1ce910398667f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-09-27 09:59:42.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/85/811085/1/check/DellEMC_Unity_FC/ea7da43/DellEMC_Unity_FC/964 : SUCCESS in 1h 10m 45s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/85/811085/1/check/DellEMC_Unity_iSCSI/378043c/DellEMC_Unity_iSCSI/831 : SUCCESS in 1h 31m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"db1c72c22168462e55be698dd4ab5fd9d4ddf1fc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-09-27 10:54:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/85/811085/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5edf504/ : TIMED_OUT in 3h 03m 24s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/85/811085/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/60c6f20/ : TIMED_OUT in 3h 03m 48s","accounts_in_message":[],"_revision_number":1},{"id":"bb2b953bad54e660276d96180bfec26279fb1472","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-09-27 11:18:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/85/811085/1/check/iscsi-driver-nimble/dc92cfe/dc92cfe.html : SUCCESS in 2h 01m 05s","accounts_in_message":[],"_revision_number":1},{"id":"c8a7ad62d356651982e4e3c3b9c6805460049920","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-27 11:57:14.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-85-811085-1/ns5_nfs : SUCCESS in 4h 7m 26s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-85-811085-1/ns5_iscsi : SUCCESS in 4h 5m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"b1e7988038f76fbbc3bf16442406986a86343201","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-09-29 04:47:49.000000000","message":"Patch Set 1: Code-Review-1\n\nPlease explain this fix in the commit msg.","accounts_in_message":[],"_revision_number":1},{"id":"5e7b7edfbaf89fb724d1ef92402543f3114e40cb","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-09-29 15:11:42.000000000","message":"Patch Set 1: Code-Review-1\n\nI also think that this should include a release note as I think this is something that an Operator might see.  Have the release note to indicate it is fixed would be good.","accounts_in_message":[],"_revision_number":1},{"id":"04cbc7759578e8a7d5dd9b13993a3e403e77a2a4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-01 12:22:29.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/85/811085/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a45732404feec2557a2476e9edd1e775688db89f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-10-03 17:16:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/85/811085/1/20211003231254/NEC-ISCSI : SUCCESS in 2h 54m 06s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/85/811085/1/20211003231254/NEC-FC : SUCCESS in 3h 03m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1cb7f3642a798f17b31fb5176e0c579cd062d99","revisions":{"a1cb7f3642a798f17b31fb5176e0c579cd062d99":{"kind":"REWORK","_number":1,"created":"2021-09-27 07:49:12.000000000","uploader":{"_account_id":33892,"name":"silphon zhang","email":"zhangxiaofan02@inspur.com","username":"zhangxiaofan02"},"ref":"refs/changes/85/811085/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/811085/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/811085/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/811085/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/811085/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/811085/1"}}},"commit":{"parents":[{"commit":"6345e39911197621417f66c9258313d598a63721","subject":"practice how to commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6345e39911197621417f66c9258313d598a63721"}]}],"author":{"name":"zhangxiaofan02","email":"zhangxiaofan02@inspur.com","date":"2021-09-27 07:47:13.000000000","tz":480},"committer":{"name":"zhangxiaofan02","email":"zhangxiaofan02@inspur.com","date":"2021-09-27 07:47:13.000000000","tz":480},"subject":"check the status of backup","message":"check the status of backup\n\nCloses-Bug: #1945169\nChange-Id: I1bb83cb271049094e35299a2bdbb54dcb6d8c29c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1cb7f3642a798f17b31fb5176e0c579cd062d99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1cb7f3642a798f17b31fb5176e0c579cd062d99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
