)]}'
{"id":"openstack%2Fcinder~813223","triplet_id":"openstack%2Fcinder~master~Iad51e8c4c78f9ed813cb44ad4e13d0d1c84cfc40","project":"openstack/cinder","branch":"master","topic":"bp/remove-sqlalchemy-migrate","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-03-08 00:18:15.000000000","reason":"Change was submitted"},"1736":{"account":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"last_update":"2022-03-08 00:18:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad51e8c4c78f9ed813cb44ad4e13d0d1c84cfc40","subject":"db: Add tests to ensure we keep migrations in sync","status":"MERGED","created":"2021-10-08 16:38:33.000000000","updated":"2022-03-08 21:40:52.000000000","submitted":"2022-03-08 00:18:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"813223-bp/remove-sqlalchemy-migrate","meta_rev_id":"4600c443f52169573fe3cfb1ff0fe62b6a47e068","_number":813223,"virtual_id_number":813223,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-08 00:18:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-03-07 17:30:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-03-03 14:50:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-02-21 10:58:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-03 14:45:56.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":1,"date":"2022-02-24 03:41:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-03 14:45:56.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-03-07 17:30:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"}],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-08 17:23:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-08 17:28:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-08 20:34:06.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-08 22:32:17.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-10-09 02:11:53.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-09 07:01:08.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-10 04:13:25.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-10-12 17:11:30.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-10-21 10:43:23.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-04 16:16:26.000000000","updated_by":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"reviewer":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"state":"CC"},{"updated":"2021-11-04 18:21:34.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-11-04 23:24:44.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-04 23:33:12.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-11-05 01:55:45.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-11-05 03:40:17.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-06 23:25:46.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-11-07 01:46:41.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-11-07 06:15:17.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-11-09 19:02:49.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-12-15 12:09:56.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2021-12-15 13:49:37.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-15 14:45:43.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-15 16:04:15.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2021-12-15 18:01:42.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-24 21:46:45.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2022-01-24 22:53:04.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-25 02:25:15.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-01-25 13:14:34.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-01-31 20:10:47.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-02-01 06:14:52.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-02-20 23:00:05.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2022-02-21 10:58:20.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-02-21 12:59:53.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2022-02-24 03:41:39.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-03-03 14:45:56.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2022-03-03 14:50:12.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"aa1aa2f02c31576e618d786ee2c94f8de9efee41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-10-08 16:38:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15fdb98416c475eaf5d8ca8eed308ace1668fe61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 17:23:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b859239efd79499cbaa0e587d640be70 : FAILURE in 26m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3b85acef5244e3ea73e42251129c880 : FAILURE in 40m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2558c363c1dc4aeb1c08bf3548a00b3d96b9a72d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 17:28:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8925aa8044a1468db514ee3c6e637a67 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bd333b4e1084a14a6845dfc4e985119 : FAILURE in 10m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3771713ad4cb4dcf829cb30caebce725 : FAILURE in 10m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52efe447d67e49cea40e114f2d04ab2c : FAILURE in 11m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cba83a170be4507addf922b4f7e6c52 : SUCCESS in 15m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ad9a30762eba4f8e993c0d9f159dcd51 : FAILURE in 12m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2b83ad8c02314254b56192a81126725c : SUCCESS in 6m 20s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6daf99f126624a9ca44d76f641a79102 : SUCCESS in 40m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ea1e122b2be4345fd58151366356ac9ced9b091","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-10-08 20:34:06.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/23/813223/1/check/DellEMC_XtremIO_FC/2fc559b/DellEMC_XtremIO_FC/12 : FAILURE in 35m 44s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/23/813223/1/check/DellEMC_XtremIO_iSCSI/beb6730/DellEMC_XtremIO_iSCSI/12 : FAILURE in 1h 51m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"161856cd057e94d7a7e7d4158c521e8243668830","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-08 22:32:17.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/23/813223/1/check/DellEMC_Unity_FC/7c8668e/DellEMC_Unity_FC/12 : FAILURE in 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/23/813223/1/check/DellEMC_Unity_iSCSI/13626e9/DellEMC_Unity_iSCSI/11 : FAILURE in 10m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"c06d7fae83a73e0550cfe229709c095429fca978","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-09 02:11:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/23/813223/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/fc183e7/fc183e7.html : TIMED_OUT in 3h 00m 46s","accounts_in_message":[],"_revision_number":1},{"id":"67e56ffbdbb334d0a2cb68bd5ceb8b2e9a9edce1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-09 05:02:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/813223/1/check/iscsi-driver-nimble/78b5f0d/78b5f0d.html : TIMED_OUT in 3h 00m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f891f07314a4a64d29fff962861578c126f76b08","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-09 07:01:08.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-813223-1/ns5_nfs : FAILURE in 1h 38m 54s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-813223-1/ns5_iscsi : FAILURE in 59m 36s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"d78c9261c9dd43430f6dca60522c7b0df50535b9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-10-10 04:13:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/813223/1/check/dsvm-tempest-cisco-zonemanager-job/3ed449a/ : FAILURE in 2h 59m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d4a1d32b294857dff8175fc924303404b755ab86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-10-12 16:01:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"5e518e92373ac0d253751d369e4ab67ab43969c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 16:29:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d8fd48e550c4963b5c08a653acbc317 : FAILURE in 18m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d151acb306d43979fda91262a256201 : FAILURE in 22m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"743803179b1eb8b2176cd3d25d700e2a8f4af24f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 17:08:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/456496abd3c5457c88671c7d507073e1 : SUCCESS in 9m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78dfd056505c4968bba753813549eea7 : FAILURE in 10m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7ee456fac1e4939bec21cf6e8866429 : FAILURE in 11m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8755a4d49f614c3fb3d24c7465a932fe : FAILURE in 12m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9931ac3453f240279304e58cf237ac2d : SUCCESS in 14m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cb466c19bbbf4039881c17d072ca57bc : FAILURE in 12m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/714ebf04b99a40a5ae962021cca667d2 : SUCCESS in 7m 23s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/792b4408c7c94a498af3a86b2bec88eb : SUCCESS in 58m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f7208531c2a379f8c89bdcd89b1d689693066631","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-10-12 17:11:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813223/2/iscsi/ : SUCCESS in 1h 01m 52s\n- cinder-smb http://cloudbase-ci.com/cinder/813223/2/smb/ : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":2},{"id":"b46ac843799de91aa9f2e7fa4f6a169b102e977b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-10-12 17:49:12.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/813223/2/check/dsvm-tempest-cisco-zonemanager-job/e2b7b83/ : FAILURE in 29m 58s","accounts_in_message":[],"_revision_number":2},{"id":"06c5a57be52f33462aa7e5b93fc7963acc9e714e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-12 18:19:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/23/813223/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/13e5148/13e5148.html : SUCCESS in 1h 17m 42s","accounts_in_message":[],"_revision_number":2},{"id":"1bee700e2ca0f82869b7bb86de30d4ab006e79d7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-12 18:22:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/23/813223/2/check/iscsi-driver-primera/d790edc/d790edc.html : SUCCESS in 1h 20m 19s","accounts_in_message":[],"_revision_number":2},{"id":"f371c6d292e0ef69cb570b219d29eaad33a83784","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-12 19:37:55.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-813223-2/ns5_nfs : FAILURE in 32m 48s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-813223-2/ns5_iscsi : FAILURE in 35m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"e8306d33c635f3c479b7d6ee7b86223b0461f81d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-12 20:46:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/23/813223/2/check/iscsi-driver-alletra-6k/1833ae1/1833ae1.html : SUCCESS in 1h 40m 18s","accounts_in_message":[],"_revision_number":2},{"id":"35569a5b68cdbd22aafc3e3508427e57ffeda90e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-12 21:00:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/813223/2/check/iscsi-driver-nimble/d1bc441/d1bc441.html : SUCCESS in 1h 48m 19s","accounts_in_message":[],"_revision_number":2},{"id":"9e65a8a5d42a6c545c55b5018f6cfe2bcffbbbf8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-12 22:14:17.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/23/813223/2/check/DellEMC_Unity_FC/0ec80dd/DellEMC_Unity_FC/31 : FAILURE in 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/23/813223/2/check/DellEMC_Unity_iSCSI/3389306/DellEMC_Unity_iSCSI/30 : FAILURE in 9m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"3678fed01e84cbfbfce913082eb2896e5730e578","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-10-21 10:43:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/23/813223/2/20211021164412/NEC-ISCSI : SUCCESS in 2h 47m 26s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/23/813223/2/20211021164412/NEC-FC : SUCCESS in 2h 59m 08s","accounts_in_message":[],"_revision_number":2},{"id":"ef765b1437fc3d2114dd5c18b6ad9a8860e78777","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-04 16:02:06.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"0db2f59b21dbc5102858b894e61bf7dec6865681","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2021-11-04 16:16:26.000000000","message":"Patch Set 3:\n\nBuild Failed. ibm-gpfs-dsvm-tempest-full: s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/100/index.html","accounts_in_message":[],"_revision_number":3},{"id":"0e6019a19d80d574cb0562651977ebda538fe1da","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-04 16:30:51.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813223/3/iscsi/ : FAILURE in 13m 56s\n- cinder-smb http://cloudbase-ci.com/cinder/813223/3/smb/ : FAILURE in 13m 58s","accounts_in_message":[],"_revision_number":3},{"id":"7e6250ed57965981c330d685da2258c0dd1b77e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 16:40:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/39d19cc7f69b409b817e9ac9c7667600 : SUCCESS in 16m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/540e45755c5442b781a760c9e9ebc81f : SUCCESS in 19m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3d3984f75a743ba91969eb771f74cbed90db59c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 17:03:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bedebfd7b8f942f7981d47de5c78857d : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52f19eb817fa4011a8a81f54d92c1093 : SUCCESS in 10m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b563d555e9f4fbc92efb8a177119826 : SUCCESS in 11m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d45afb3629dc48b19ddf10628eb72d3a : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f79cb8e9b1144df0a78d1890dae3390c : SUCCESS in 14m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f701c1a8ac964ff58da83cf757c1ab53 : SUCCESS in 15m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5dc925aa20bf477abc49819932b70114 : SUCCESS in 7m 07s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0b0015bb1fed40b3940d0655c6de3c90 : FAILURE in 32m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"59812fb5053c8dae41b9e341331a92b67d6d81fb","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-11-04 18:21:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/23/813223/3/upstream-solidfire-check/cinder-solidfire-iSCSI/e820cd8/ : SUCCESS in 1h 15m 19s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"571b8bd692f2de03b38478f6e5e9aeb692dc7b4a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-04 19:24:18.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/23/813223/3/check/iscsi-driver-primera/cfcf121/cfcf121.html : FAILURE in 1h 24m 17s","accounts_in_message":[],"_revision_number":3},{"id":"6c1b2d1075381a482e6d7d0bbbaa405ccc4ba561","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-04 20:59:01.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/23/813223/3/check/iscsi-driver-alletra-6k/65676e8/65676e8.html : SUCCESS in 1h 39m 12s","accounts_in_message":[],"_revision_number":3},{"id":"cc55d1dda8e38fcd6d9d9bc19cb237d5591bd447","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-04 22:34:06.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/813223/3/check/iscsi-driver-nimble/5f7fd3a/5f7fd3a.html : FAILURE in 1h 13m 01s","accounts_in_message":[],"_revision_number":3},{"id":"bf20a7f07bef5a496f47a0a8a61aa5b5c1c043de","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-04 23:10:08.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/23/813223/3/check/dsvm-tempest-cisco-zonemanager-job/dcf4a43/ : SUCCESS in 1h 08m 28s","accounts_in_message":[],"_revision_number":3},{"id":"a9d1db0c23fb9d3caec0619e1b3b7fbd8e0aa3ac","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-11-04 23:24:44.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/23/813223/3/check/DellEMC_PowerFlex/dcd1f7d/DellEMC_PowerFlex/73 : FAILURE in 1h 53m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"8a0b8c8291eecc4d273a7ec098936a8e82f95319","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-11-04 23:33:12.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/23/813223/3/check/DellEMC_PowerStore_iSCSI/c3fe207/DellEMC_PowerStore_iSCSI/34 : SUCCESS in 1h 49m 20s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/23/813223/3/check/DellEMC_PowerStore_FC/215e2f6/DellEMC_PowerStore_FC/41 : SUCCESS in 1h 19m 20s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"73c24e02c21df8a4191b86b596fc9867711c6ca7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-05 00:35:48.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-813223-3/ns5_nfs : SUCCESS in 4h 26m 39s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-813223-3/ns5_iscsi : SUCCESS in 4h 30m 58s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"4051e70f42e49b00fe6a07e7ef4ae5610f56f837","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-11-05 01:55:45.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI http://os-logs.tristero.net/dell-ci/813223-3/PVME-iSCSI : SUCCESS in 58:51.61\n- PVME-FC http://os-logs.tristero.net/dell-ci/813223-3/PVME-FC : SUCCESS in 59:29.96","accounts_in_message":[],"_revision_number":3},{"id":"4d249d4fef7739127355cc868c86408249e702a7","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-11-05 03:40:17.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/23/813223/3/check/DellEMC_VNX_iSCSI/6b47801/DellEMC_VNX_iSCSI/70 : FAILURE in 2h 56m 26s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/23/813223/3/check/DellEMC_VNX_FC/d6a7249/DellEMC_VNX_FC/69 : SUCCESS in 2h 31m 15s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"c0fc917f94cd1fab5ae76d21286655953c3c86ee","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-05 17:47:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/23/813223/3/20211105235530/NEC-ISCSI : SUCCESS in 2h 49m 37s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/23/813223/3/20211105235530/NEC-FC : SUCCESS in 2h 51m 31s","accounts_in_message":[],"_revision_number":3},{"id":"64bba40a395cad7c08cef0d6d13a282571cc48fe","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-11-06 23:25:46.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/813223-3/Seagate-iSCSI : SUCCESS in 1:21:30\n- Seagate-FC http://os-logs.tristero.net/seagate-ci/813223-3/Seagate-FC : SUCCESS in 49:11.76","accounts_in_message":[],"_revision_number":3},{"id":"11f7b28aba9523368cca70227a8eb1fd0cefea2d","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-11-07 01:46:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/813223-3/HPMSA-iSCSI : SUCCESS in 1:11:22\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/813223-3/HPMSA-FC : SUCCESS in 42:13.16","accounts_in_message":[],"_revision_number":3},{"id":"24e916583ed7416bd2033eeb09e19b54381f76aa","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-11-07 06:15:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/813223-3/Lenovo-iSCSI : SUCCESS in 1:25:15\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/813223-3/Lenovo-FC : SUCCESS in 2:31:55","accounts_in_message":[],"_revision_number":3},{"id":"861342d8c12c3eb69fe60595d32fb1a3063bf98f","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-11-09 19:02:49.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/23/813223/3/check/dsvm-tempest-TOYOU-ACS5000-driver/473cd20 : FAILURE in 2m 11s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/23/813223/3/check/fc-test/147a29d : FAILURE in 39m 55s","accounts_in_message":[],"_revision_number":3},{"id":"c270e6573e3f66efd0d14e88767fa2a0ad9fa8b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-15 10:47:01.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"0d3a823d2f591a78a322006933c1f80d1ada50c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 11:22:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7af4b1bbf8f421db94b8a0d1554c300 : SUCCESS in 19m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cebe011196194b7c958bca7d5ac4743c : SUCCESS in 18m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"782d596c4124abbb4d05f9b311d51ec24854fb19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 11:36:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0117314588bf46a88e3eec7d12087334 : SUCCESS in 10m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a26519734c7c45bcb8c18df6622b5e6e : SUCCESS in 11m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3fe30f5c06c4914bdc3ae563d6eb2a1 : SUCCESS in 13m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/146a71f2980a470bb714410cfad53b74 : SUCCESS in 15m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ceaef9fd02c842be89cbdc3ae189d3f1 : SUCCESS in 16m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/61299ff1eea74cbc95bc0c8c498cf416 : SUCCESS in 8m 34s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/720ccba12f724be5b3cf8fb64e2a0ef6 : SUCCESS in 41m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"930c6e27114e47d4596cdb110e9eb62ec0a5fd3b","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-12-15 11:40:54.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"c03bf9f932a5f74a9b7b7ce0031e020c39e3f93c","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-12-15 12:09:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-23-813223-4/1639568211 : SUCCESS in 29m 48s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"82fbc114321c2d9d886b22a8727d36c7a3d19eed","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-12-15 12:44:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/23/813223/4/upstream-solidfire-check/cinder-solidfire-iSCSI/7e86b2a/ : SUCCESS in 1h 04m 02s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"402651641305baea0d2aa6c9c4afd1d2bd5aa5f0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-15 12:51:01.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813223/4/iscsi/ : SUCCESS in 56m 40s\n- cinder-smb http://cloudbase-ci.com/cinder/813223/4/smb/ : SUCCESS in 1h 08m 54s","accounts_in_message":[],"_revision_number":4},{"id":"3b1925b869e118c72d1bd2d9da2bf08451a3f898","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-12-15 13:49:37.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/23/813223/4/ : SUCCESS in 1h 21m 29s","accounts_in_message":[],"_revision_number":4},{"id":"e08aefbfbef9083258f8a80aca2a1ed26905f933","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-12-15 14:45:43.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/23/813223/4/upstream-check/cinder-cDOT-iSCSI/6eecedc/ : SUCCESS in 34m 10s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/23/813223/4/upstream-check/cinder-cDOT-NFS/c1c110f/ : FAILURE in 2h 44m 21s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/23/813223/4/upstream-check/cinder-cDOT-FCP/5197677/ : FAILURE in 3h 07m 42s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"5d38ed34669e77d93d8c32d5fe3cbede343efc78","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-15 16:04:15.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/23/813223/4/check/iscsi-driver-alletra-6k/e1577fa/e1577fa.html : FAILURE in 1h 02m 43s","accounts_in_message":[],"_revision_number":4},{"id":"f25c761eac2e35a9acef9f476e57fe508e3fe4a8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-15 17:40:15.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-813223-4/ns5_nfs : FAILURE in 4h 31m 57s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-813223-4/ns5_iscsi : SUCCESS in 4h 37m 28s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"6a4e35e7872d154caef7ce47e841d06422c62177","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-15 18:01:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/813223/4/check/iscsi-driver-nimble/af7e000/af7e000.html : SUCCESS in 1h 49m 51s","accounts_in_message":[],"_revision_number":4},{"id":"9a065d37af1ec867be0a3744f9d109ede56c5636","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-15 18:01:42.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bb573d9d152649e3b2541be9867529e2 : SUCCESS in 1h 22m 28s","accounts_in_message":[],"_revision_number":4},{"id":"bd9ff2eb108b0167f138f14f94cb9ef22641aded","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-15 21:11:01.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/23/813223/4/check/dsvm-tempest-TOYOU-ACS5000-driver/89aa981 : FAILURE in 1h 29m 15s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/23/813223/4/check/fc-test/bffad89 : FAILURE in 1h 18m 39s","accounts_in_message":[],"_revision_number":4},{"id":"568ba85a6bf3387927557d5565462e398222f361","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-12-16 21:28:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/23/813223/4/20211217003127/NEC-FC : SUCCESS in 2h 50m 42s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/23/813223/4/20211217003127/NEC-ISCSI : SUCCESS in 5h 56m 56s","accounts_in_message":[],"_revision_number":4},{"id":"9972af9606556ff40878368d9135ef91b27d8372","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-01-24 21:46:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a1f82fff02e3fe4cc7ae36c4cce285ebe59bd0d8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-01-24 21:48:49.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"72b0108fd8be92e16102fbdcc74d203581731cf3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 22:13:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f38c99b8e27249d094b3e951f39feee2 : FAILURE in 22m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5504d4d698dc4e3bb30b95e1d8b0afdc : FAILURE in 22m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"34b92cfed96c61f7fe61c2a4f0285785637bb4d4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-24 22:14:41.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/23/813223/4/upstream-check/cinder-cDOT-iSCSI/d0fdeb8/ : NOT_REGISTERED\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/23/813223/4/upstream-check/cinder-cDOT-NFS/96a25e7/ : NOT_REGISTERED\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/23/813223/4/upstream-check/cinder-cDOT-FCP/7e26f84/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"b27c42a116976daa6a4d90e5a919bc6134c93c1c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-01-24 22:15:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/23/813223/4/upstream-solidfire-check/cinder-solidfire-iSCSI/fd1ebb1/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"bb6e46f591a5b98e23bb23905c4191826edb21a7","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-01-24 22:30:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-23-813223-4/1643062380 : SUCCESS in 17m 14s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"5d1b21966d8cefd68f2b8358a0c626164e9362f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 22:36:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/204e1d998bdc4584b2682ebd553ca752 : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88d0266db3984cd7b7451f0d590b7c20 : SUCCESS in 10m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74d1338a7de34109aa6a5221d40a383e : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c7b5b32ca06412e9d4b14904f79c8df : SUCCESS in 12m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8b6c93067702400384eb2503b86fdbb3 : SUCCESS in 14m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3c72df709272464f9ad8c05443048d9e : SUCCESS in 6m 42s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/00fb3f4119e04782947cd9deda9b6c52 : SUCCESS in 42m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"482bf7748d14dc23fa7ee79e586a41d2d99a7ab9","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-24 22:37:16.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/23/813223/4/upstream-check/cinder-cDOT-iSCSI/323269b/ : NOT_REGISTERED\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/23/813223/4/upstream-check/cinder-cDOT-NFS/ee9497f/ : NOT_REGISTERED\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/23/813223/4/upstream-check/cinder-cDOT-FCP/1b1f11f/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"27027fe57552fbd3f1ba15c13df0d440acd5a0ae","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-01-24 22:38:08.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/23/813223/4/upstream-solidfire-check/cinder-solidfire-iSCSI/20d5a00/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"c587e8228f213b6ccdb8c6b13ee755e26c252b6d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-24 22:51:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813223/4/iscsi/ : SUCCESS in 55m 17s\n- cinder-smb http://cloudbase-ci.com/cinder/813223/4/smb/ : SUCCESS in 1h 04m 00s","accounts_in_message":[],"_revision_number":4},{"id":"9b8127693b5f97c9e739d8facf29236dfc6a0a1f","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-01-24 22:53:04.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/23/813223/4/check/DellEMC_PowerMAX_FC/f88e10b/DellEMC_PowerMAX_FC/221 : FAILURE in 37m 28s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/23/813223/4/check/DellEMC_PowerMAX_iSCSI/b8e2da9/DellEMC_PowerMAX_iSCSI/234 : FAILURE in 39m 27s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"3777821e68ad81b9f8a48d4f39be19e5b0060319","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-24 23:34:35.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/23/813223/4/check/DellEMC_PowerFlex/50d2f0d/DellEMC_PowerFlex/308 : SUCCESS in 1h 20m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"6886c1e811160bce126e2fee8839269296a7064f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-01-24 23:42:05.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/23/813223/4/check/DellEMC_PowerStore_iSCSI/0b70142/DellEMC_PowerStore_iSCSI/277 : SUCCESS in 1h 28m 29s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/23/813223/4/check/DellEMC_PowerStore_FC/1b146b8/DellEMC_PowerStore_FC/280 : SUCCESS in 1h 15m 28s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"fd388d500ee0c36b4698606d24d3eb5e54d90b31","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-24 23:46:56.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/23/813223/4/check/DellEMC_Unity_FC/c5f46bf/DellEMC_Unity_FC/471 : SUCCESS in 1h 03m 02s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/23/813223/4/check/DellEMC_Unity_iSCSI/42856f6/DellEMC_Unity_iSCSI/466 : SUCCESS in 1h 33m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"13c574bb143e650b87aeb3518a92c45ec1ef9603","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-01-25 02:02:27.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/23/813223/4/check/DellEMC_XtremIO_FC/a221a4e/DellEMC_XtremIO_FC/553 : SUCCESS in 56m 18s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/23/813223/4/check/DellEMC_XtremIO_iSCSI/9f5803f/DellEMC_XtremIO_iSCSI/563 : SUCCESS in 1h 27m 53s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"2b52fc06d62329f282d2a3908e8f08e280f8ecf7","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-01-25 02:25:15.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/23/813223/4/check/DellEMC_SC_FC/c2909f0/DellEMC_SC_FC/244 : SUCCESS in 46m 18s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/23/813223/4/check/DellEMC_SC_iSCSI/c1be90e/DellEMC_SC_iSCSI/242 : FAILURE in 2h 10m 45s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"2038cbb18344f9b604a322c3bc9553f0d63ab2e0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-01-25 13:14:34.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"46d4c6b5b9370ab6844d7550c1beee8f7a3f049f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-31 18:17:06.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"92cb9e4c4a329e3706ef4f08b15d015f4d151176","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-01-31 18:21:10.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"cde5ed2bd02793c07c9910cdf45947e80855c93b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-31 19:36:01.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813223/5/iscsi/ : SUCCESS in 1h 17m 29s\n- cinder-smb http://cloudbase-ci.com/cinder/813223/5/smb/ : SUCCESS in 1h 15m 16s","accounts_in_message":[],"_revision_number":5},{"id":"4fd399b16dbaeba8d0c6185eeb020f53ba11d4b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 20:03:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38a25814aeeb482c9a98fe1752a1ce39 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c7dda9092e1437c8ee4240c3ed9b02c : SUCCESS in 11m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6330136163e4fae905356d27b08eb68 : SUCCESS in 10m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e59ed39c269d462fa7d1827bd760e9c4 : SUCCESS in 11m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bf94d62e47934894af72326a08c8ca46 : SUCCESS in 15m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a2ac91b16cbe42fba2a50d3869e652a3 : SUCCESS in 7m 58s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5bd9e5e740e44531b781c03201c214b3 : SUCCESS in 1h 06m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"90bf35b31480bbe0921ff6d621ae1c6b98cdfcbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 20:04:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6d48f5c35594a6ebfff971f593536e1 : FAILURE in 19m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd9621d17aa74e3997ac8931e11ee44e : FAILURE in 37m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b49f78a4236cc8011f13cdd2ed90b211972c119a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-31 20:10:47.000000000","message":"Patch Set 5:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/31_01_2022_20:03:20\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"0cae1c84dd00a4354e032a7db7251170f69334db","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-31 20:30:44.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/23/813223/5/check/DellEMC_PowerFlex/e80a4a8/DellEMC_PowerFlex/343 : FAILURE in 27m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"844df98e87e950d7539cb778c527cf47c6265be8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-01-31 20:50:11.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/23/813223/5/check/iscsi-driver-alletra-6k/6bd73bb/6bd73bb.html : SUCCESS in 1h 37m 24s","accounts_in_message":[],"_revision_number":5},{"id":"26e5a310a902dd5d69791f6d57c0633d8c046ccf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-01-31 21:00:05.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/23/813223/5/check/iscsi-driver-primera/8222680/8222680.html : SUCCESS in 1h 24m 31s","accounts_in_message":[],"_revision_number":5},{"id":"b872d08a0d96a7ae1f11883df6ad3e527378f9a4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-31 21:20:15.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/23/813223/5/upstream-check/cinder-cDOT-iSCSI/a3ee6e7/ : SUCCESS in 35m 15s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/23/813223/5/upstream-check/cinder-cDOT-NFS/dc4b039/ : SUCCESS in 36m 20s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/23/813223/5/upstream-check/cinder-cDOT-FCP/2703fa9/ : FAILURE in 12m 58s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"f302ce1167d358d2eeb1c75c87a267187f3348ce","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-01-31 21:24:29.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/23/813223/5/upstream-solidfire-check/cinder-solidfire-iSCSI/1834352/ : SUCCESS in 1h 04m 41s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"0bddc87a296bf9eeb4a8744ca5717fe28d748eed","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-01-31 23:33:00.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-23-813223-5/1643659394 : SUCCESS in 43m 42s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":5},{"id":"41ad4b38fed0e885850c1fc8a96b63aae0cd5272","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-01 00:02:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-23-813223-5/1643659489 : SUCCESS in 29m 39s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":5},{"id":"3e38d6543a7faf6d5bcbd241b1449d59e261dcce","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-01 00:08:43.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/23/813223/5/check/dsvm-tempest-cisco-zonemanager-job/549d5d9/ : FAILURE in 1h 08m 22s","accounts_in_message":[],"_revision_number":5},{"id":"f327c460cd34908cc33516a3af4eefb60148a5e9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-01 00:36:27.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-813223-5/ns5_nfs : SUCCESS in 4h 41m 4s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-813223-5/ns5_iscsi : SUCCESS in 4h 48m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"1f50a40154b2ac98c3851282b8149bbbea901308","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-01 00:56:11.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/23/813223/5/check/iscsi-driver-nimble/adb88cc/adb88cc.html : SUCCESS in 1h 50m 06s","accounts_in_message":[],"_revision_number":5},{"id":"ebde39f249f0b50788e269b5080da24424194bf6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-02-01 01:41:50.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/23/813223/5/check/DellEMC_Unity_FC/01c8836/DellEMC_Unity_FC/492 : FAILURE in 1h 19m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/23/813223/5/check/DellEMC_Unity_iSCSI/335d798/DellEMC_Unity_iSCSI/486 : SUCCESS in 1h 24m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"ce7dea3d4ea359e78af9f3bf09f8fe52e6c7940a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-02-01 01:59:42.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/23/813223/5/check/DellEMC_PowerMAX_FC/63127d8/DellEMC_PowerMAX_FC/256 : FAILURE in 52m 57s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/23/813223/5/check/DellEMC_PowerMAX_iSCSI/4382b75/DellEMC_PowerMAX_iSCSI/269 : FAILURE in 53m 44s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"982b001a1567c02a01041408437def38051cd828","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-02-01 02:06:08.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/23/813223/5/check/DellEMC_PowerStore_iSCSI/fce3039/DellEMC_PowerStore_iSCSI/294 : FAILURE in 1h 20m 30s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/23/813223/5/check/DellEMC_PowerStore_FC/fe8e9ac/DellEMC_PowerStore_FC/311 : SUCCESS in 1h 09m 37s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"9d072b906b12ca364c1958829489915fb4ce8d10","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-01 04:33:14.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/23/813223/5/check/DellEMC_XtremIO_FC/ecb6c8c/DellEMC_XtremIO_FC/606 : SUCCESS in 53m 35s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/23/813223/5/check/DellEMC_XtremIO_iSCSI/8e2e22e/DellEMC_XtremIO_iSCSI/616 : SUCCESS in 1h 23m 38s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"eecb3bb129b6b67eb839728346e1793674fef701","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-02-01 06:14:52.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/23/813223/5/20220201124507/NEC-V-FC : SUCCESS in 2h 17m 19s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/23/813223/5/20220201124507/NEC-V-ISCSI : SUCCESS in 2h 29m 42s","accounts_in_message":[],"_revision_number":5},{"id":"5810a089b44fede3cb339eb665111ba1b7682cf8","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-02-01 08:24:43.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/23/813223/5/check/DellEMC_SC_FC/60b5469/DellEMC_SC_FC/279 : FAILURE in 1h 07m 10s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/23/813223/5/check/DellEMC_SC_iSCSI/dc64bc9/DellEMC_SC_iSCSI/275 : FAILURE in 1h 39m 44s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"a1daa2c202378a992b7f4375842cb457a3377ca7","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-01 10:52:55.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/813223-5/HPMSA-iSCSI/ : SUCCESS in 51:23.48\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/813223-5/HPMSA-FC/ : SUCCESS in 43:20.52","accounts_in_message":[],"_revision_number":5},{"id":"43dc8a1ccbedecea656d5272da54ef0e3cfce415","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-01 14:06:08.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/813223-5/Lenovo-iSCSI/ : SUCCESS in 59:11.58\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/813223-5/Lenovo-FC/ : SUCCESS in 56:51.40","accounts_in_message":[],"_revision_number":5},{"id":"777367c50e2a81a27e8cdfa14f6f3c7bffa1fc60","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-01 14:58:24.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/813223-5/PVME-iSCSI/ : SUCCESS in 1:04:23\n- PVME-FC https://os-logs.seagate.com/dell-ci/813223-5/PVME-FC/ : SUCCESS in 1:00:48","accounts_in_message":[],"_revision_number":5},{"id":"e5c9dc3e8e8b2fc59bce14ef847f1fb1d6d02239","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-02-01 15:41:12.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/813223-5/Seagate-iSCSI/ : SUCCESS in 1:06:18\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/813223-5/Seagate-FC/ : SUCCESS in 1:02:15","accounts_in_message":[],"_revision_number":5},{"id":"a75eb0ffff66ed51e5a44bc433c81eec6f57df05","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-02 16:18:22.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/23/813223/5/ : SUCCESS in 1h 52m 58s","accounts_in_message":[],"_revision_number":5},{"id":"c8173125090d1f464267fc6d575862c2a97972ec","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-02-04 01:36:56.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/23/813223/5/20220204071350/NEC-FC : SUCCESS in 3h 21m 56s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/23/813223/5/20220204071350/NEC-ISCSI : SUCCESS in 3h 23m 03s","accounts_in_message":[],"_revision_number":5},{"id":"3e11fa23aed3197a7311e96c81fdaa2fc46e1200","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-20 19:58:21.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"8d1dd18bfe48ba3c6c033c45aa177c50362d5efa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 20:17:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f37fd7f01ce84f08a6a897aeca2f34ec : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ec2ee60a7614b3ea61f8225758e6ac4 : SUCCESS in 13m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a03e0e0f49846b684e3fa69e184fe5f : SUCCESS in 11m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f18591a616074e54a900e679421c045e : SUCCESS in 13m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/57a1b1b052cf4a0db31f0efbfafb7b14 : SUCCESS in 11m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/85812f7821fd4b178219b9f726ca80bd : SUCCESS in 6m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fb8c4f6715f3a774d2b85c59a815814c3b3076f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 20:29:49.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/20c486518ead4404bd3110b3aa4221ad : FAILURE in 22m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e22b28fe50db4945b1d7bce9c4a46db6 : FAILURE in 24m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1bf5b76c3098b23007eea759fee699ab94966838","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-20 20:35:10.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-23-813223-6/1645388256 : SUCCESS in 17m 32s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":6},{"id":"ef450f6917288e2b98ccbb954b9d75ce614346f7","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-20 21:08:42.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-23-813223-6/1645388989 : SUCCESS in 17m 03s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":6},{"id":"ed9ce65a234a83c75633ae188ebab458df64e803","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2022-02-20 23:00:05.000000000","message":"Patch Set 6:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/23/813223/6/fc : FAILURE in 0h 24m 55s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/23/813223/6/iscsi : SUCCESS in 1h 10m 46s","accounts_in_message":[],"_revision_number":6},{"id":"3fab8c54b6600cab7c0c89837d515f05e0f81cb3","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-20 23:29:41.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/813223-6/HPMSA-iSCSI/ : SUCCESS in 49:58.03\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/813223-6/HPMSA-FC/ : SUCCESS in 42:44.91","accounts_in_message":[],"_revision_number":6},{"id":"dc24ad3d75131e5e497bb48466f8aa7854d65976","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-21 00:16:25.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813223/6/iscsi/ : SUCCESS in 55m 08s\n- cinder-smb http://cloudbase-ci.com/cinder/813223/6/smb/ : SUCCESS in 1h 07m 28s","accounts_in_message":[],"_revision_number":6},{"id":"11902fa7dfd516f6ff9d6d79a6fb0d9edc263583","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-21 00:25:26.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/813223-6/Lenovo-iSCSI/ : SUCCESS in 1:03:48\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/813223-6/Lenovo-FC/ : SUCCESS in 56:35.73","accounts_in_message":[],"_revision_number":6},{"id":"1224fde5f56b8c03def59f07aa2d6fd1443472b7","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-21 00:30:21.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/813223-6/PVME-iSCSI/ : SUCCESS in 1:03:28\n- PVME-FC https://os-logs.seagate.com/dell-ci/813223-6/PVME-FC/ : SUCCESS in 1:01:05","accounts_in_message":[],"_revision_number":6},{"id":"5a42f9d283968f199f6e0e6b37c524804cb7f401","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-02-21 00:44:29.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/813223-6/Seagate-iSCSI/ : SUCCESS in 1:05:19\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/813223-6/Seagate-FC/ : SUCCESS in 1:04:42","accounts_in_message":[],"_revision_number":6},{"id":"0fbf3297928401d0df9a09b427e447d32ce71cf0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-21 07:12:51.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/23/813223/6/check/dsvm-tempest-cisco-zonemanager-job/7452ce3/ : SUCCESS in 1h 08m 41s","accounts_in_message":[],"_revision_number":6},{"id":"a730b39dbc22ccaf68501907eb4c1d4bd6105688","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-21 09:44:19.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/23/813223/6/check/iscsi-driver-alletra-6k/2645a68/2645a68.html : SUCCESS in 1h 38m 28s","accounts_in_message":[],"_revision_number":6},{"id":"72978cec7b30be7c3f7df013e8ee6aeceda24de0","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-02-21 10:58:20.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"b19679e230b969f88a4ad2ce720ae48ebcbb5c56","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-21 12:49:02.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-813223-6/ns5_nfs : SUCCESS in 5h 35m 29s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-23-813223-6/ns5_iscsi : SUCCESS in 5h 8m 35s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"b26ae2136e81d45bc6239ac0e826073babbe3b31","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-02-21 22:16:45.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/23/813223/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/cc731f2/cc731f2.html : SUCCESS in 1h 24m 50s","accounts_in_message":[],"_revision_number":6},{"id":"2cb6fd2b351610949c0b93e430b72327fadab0ea","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-22 22:30:58.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/23/813223/6/ : SUCCESS in 1h 48m 29s","accounts_in_message":[],"_revision_number":6},{"id":"3f3f006880f6945dd49ab20a1325bf3544411f05","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-02-24 02:00:44.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/23/813223/6/20220224074236/NEC-ISCSI : SUCCESS in 3h 17m 27s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/23/813223/6/20220224074236/NEC-FC : SUCCESS in 3h 18m 05s","accounts_in_message":[],"_revision_number":6},{"id":"b18984f3a382c23f1543ea84944279783f71304e","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-02-24 03:41:39.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"52245dfadf2bcfc40a9c8321cdc15ca7c3b5fa2b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-02-25 13:23:21.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/23/813223/6/20220225170025/NEC-V-ISCSI : SUCCESS in 2h 37m 38s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/23/813223/6/20220225170025/NEC-V-FC : SUCCESS in 5h 22m 53s","accounts_in_message":[],"_revision_number":6},{"id":"65449a476f3a40ed3aede2bbf5d1e898daa60eb0","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-03-03 14:45:56.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fe0e020a1428cc4ad86793648022ae718f124168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 14:46:34.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8a9f568810bfe79d312640e173510e8ba59e97d4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-03-03 14:50:12.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"1dd6f019ddc0fe9c202b8a48b2bdb670f7d55b42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 15:14:07.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"95996369a9632bae8d8bd18b4692d65249201afc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-07 17:30:47.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"96e64642355c319365703ef7962647668a751f55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 18:00:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a14fa37caa4b473fa83940ce909b9237 : FAILURE in 22m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e0e45b646fef44d5a9793caea4f17d68 : FAILURE in 27m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4c19811db4a257eac2a52703f34ae5c5cc89e4e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 18:33:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/303960b168874ee29c941a6b0c94db9d : SUCCESS in 9m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d26c308eae8b4bf5a4470f034dcaac7c : SUCCESS in 11m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65b2a439b45742a6bd455c135d1d4196 : SUCCESS in 12m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d7c165ac07492185c26cbea2f41c47 : SUCCESS in 13m 02s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6fd60109178f463eb6790def5c882534 : SUCCESS in 1h 00m 36s (non-voting)\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d69c61af5c2540c1a3e2f85a9e8a6df3 : SUCCESS in 15m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e3d1d6bce8d14c8695e057e09168f0ca : SUCCESS in 8m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b02d17b7df5e870a8c2896a56b7ce20852c731d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 18:34:04.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"bf68fc2b78f340582b7babcf5d9bcd2273c3b34c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-03-07 18:42:56.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813223/6/iscsi/ : SUCCESS in 1h 11m 42s\n- cinder-smb http://cloudbase-ci.com/cinder/813223/6/smb/ : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":6},{"id":"4cb0de4cb9ccf2d415be7e6ee4d8cc4e09f6e157","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 19:42:44.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0e0a01e74664d22aa21184e1b41d9ec : SUCCESS in 9m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8362038991784ad58a42589309f00dc0 : FAILURE in 17m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0745b7d807ec49bf89199bcdeaff2353 : SUCCESS in 14m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c36ff6587914d2da50de01a790eb3ee : SUCCESS in 15m 17s","accounts_in_message":[],"_revision_number":6},{"id":"b00017d2b2dfbbd5d51580005c01d1c0a28829b4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-03-07 21:01:58.000000000","message":"Patch Set 6:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/07_03_2022_20_24_16\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"2d34ab58018dfae201f07a61134147ccf6f3e56a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-07 22:28:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cd7e8282f595a7f9153443b116a5c23932199005","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 22:51:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e510f1c2936840e08235df67b0804e52 : SUCCESS in 19m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2573c1b3c47b4fd1931f4bcc67e1cde0 : SUCCESS in 21m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"afdfe154ad7582c0fda82bf16370d1e5ffc7224c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-03-07 23:36:29.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813223/6/iscsi/ : SUCCESS in 1h 07m 56s\n- cinder-smb http://cloudbase-ci.com/cinder/813223/6/smb/ : SUCCESS in 57m 19s","accounts_in_message":[],"_revision_number":6},{"id":"dec4dbaf5fd7397c2c3f9cb4742164cf1c93cba0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 23:44:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa7478bc51784625b622cefe6fd6d9df : SUCCESS in 8m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59a560d4212040888ee001e172f9b1df : SUCCESS in 13m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60ef92fb0e614db6865e4bc170d8de46 : SUCCESS in 13m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4505eb0722c4a8e82c0decba03a0b1b : SUCCESS in 12m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c3c9bfe954d3435db75059b0526820c4 : SUCCESS in 1h 10m 01s (non-voting)\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a5b346c91360452d9ee7dd509cd6b8f8 : SUCCESS in 17m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/13e730769bf14c1d8453023d0d20b159 : SUCCESS in 6m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"93bb3d71b0b9976145c8c3d03473712ca1d55445","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 23:45:40.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1c8571efc4661461c4e2ced37ef33304d0f19861","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 00:18:15.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ed342e1bae94cf199dd01c062eae2fa : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f82e4d0483fc4f9899de2b22c881330a : SUCCESS in 11m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3262e4233c944065b7277b07da0147c9 : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a52394dae0a44f109b477c59d3a1ff31 : SUCCESS in 11m 48s","accounts_in_message":[],"_revision_number":6},{"id":"da0778418559c1eaa5b5301cf93d9dae36289481","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 00:18:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"0713e11339852c435c9c48a545e0fbb1fe1de162","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 00:19:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c90a03939e148db9286df41b9d7737c : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":6},{"id":"70c21af17fd0023dc5c6fb04ab4d5f908dd07307","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-03-08 00:50:49.000000000","message":"Patch Set 6:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/08_03_2022_00_13_16\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"3606d2b3e94b90cdd1d5059ee85b92bb156a38d3","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-03-08 04:50:12.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/813223-6/HPMSA-iSCSI/ : SUCCESS in 51:23.52\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/813223-6/HPMSA-FC/ : SUCCESS in 45:45.60","accounts_in_message":[],"_revision_number":6},{"id":"64c21bd410e1c7e4c6f68560f09fa748eb81f986","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-03-08 05:00:45.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/23/813223/6/check/dsvm-tempest-cisco-zonemanager-job/36121ba/ : SUCCESS in 1h 17m 35s","accounts_in_message":[],"_revision_number":6},{"id":"56f3c90e9c913cc88f6dd949c0d1e6125a62edaf","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-03-08 07:41:33.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/813223-6/Lenovo-iSCSI/ : SUCCESS in 1:04:42\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/813223-6/Lenovo-FC/ : SUCCESS in 1:03:17","accounts_in_message":[],"_revision_number":6},{"id":"41a8e8ff10be1cc5cab2ca71c6e4bb3d9f96465b","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-03-08 07:48:48.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/813223-6/PVME-iSCSI/ : SUCCESS in 1:04:41\n- PVME-FC https://os-logs.seagate.com/dell-ci/813223-6/PVME-FC/ : SUCCESS in 1:03:38","accounts_in_message":[],"_revision_number":6},{"id":"a2f02890f2a527b9c78df0e346748265fb9e4ddd","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-03-08 08:17:51.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/813223-6/Seagate-iSCSI/ : SUCCESS in 1:05:30\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/813223-6/Seagate-FC/ : SUCCESS in 1:04:08","accounts_in_message":[],"_revision_number":6},{"id":"33b654d4d2cbf7c3e9f99d76bc6dd52cd6a74b64","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-03-08 13:24:12.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/813223-6/HPMSA-iSCSI/ : SUCCESS in 52:29.52\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/813223-6/HPMSA-FC/ : SUCCESS in 45:37.17","accounts_in_message":[],"_revision_number":6},{"id":"9855e03902cffd3ee95e9f43aa826b0e0d6c60d1","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2022-03-08 16:00:07.000000000","message":"Patch Set 6:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/23/813223/6/fc : FAILURE in 0h 24m 55s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/23/813223/6/iscsi : SUCCESS in 1h 10m 46s","accounts_in_message":[],"_revision_number":6},{"id":"1ab7fd2492dafb954b20292280010346f9abab73","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-03-08 17:58:13.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/813223-6/Lenovo-iSCSI/ : SUCCESS in 1:03:14\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/813223-6/Lenovo-FC/ : SUCCESS in 59:11.12","accounts_in_message":[],"_revision_number":6},{"id":"96445fc435c22150a7a57c8eaad87d285f1a332d","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-03-08 18:17:36.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/813223-6/PVME-iSCSI/ : SUCCESS in 1:03:46\n- PVME-FC https://os-logs.seagate.com/dell-ci/813223-6/PVME-FC/ : SUCCESS in 1:01:48","accounts_in_message":[],"_revision_number":6},{"id":"b87e95fdefdc7cffd3034ec5bfcf12f4fdae95d3","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-03-08 19:06:55.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/813223-6/Seagate-iSCSI/ : SUCCESS in 1:04:30\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/813223-6/Seagate-FC/ : SUCCESS in 1:02:10","accounts_in_message":[],"_revision_number":6},{"id":"4600c443f52169573fe3cfb1ff0fe62b6a47e068","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-03-08 21:40:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"8b5705156ca22a64b65b61d17cd64ff39c7aeb91","revisions":{"cd59eb7c1c660da842b7ec73d984a43170cfaf76":{"kind":"REWORK","_number":1,"created":"2021-10-08 16:38:33.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/813223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/813223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/813223/1"}}},"commit":{"parents":[{"commit":"240dbf423163f38d605a4ceb7b6b63274112a6dc","subject":"db: Resolve additional migration-model mismatches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/240dbf423163f38d605a4ceb7b6b63274112a6dc"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-10-07 16:40:13.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-10-08 16:38:26.000000000","tz":60},"subject":"db: Add tests to ensure we keep migrations in sync","message":"db: Add tests to ensure we keep migrations in sync\n\nOur database migrations were previously out of sync with the models\nused. These issues are now resolved, but we want to prevent this\nhappening again. Add tests to do just that. oslo.db does most of the\nheavy lifting here and we simply use the tests that this provides.\n\nChange-Id: Iad51e8c4c78f9ed813cb44ad4e13d0d1c84cfc40\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd59eb7c1c660da842b7ec73d984a43170cfaf76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd59eb7c1c660da842b7ec73d984a43170cfaf76"}]},"branch":"refs/heads/master"},"ef7b96b557b235c4d351bf297870eccf9598a639":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-12 16:01:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/813223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/813223/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/813223/2"}}},"commit":{"parents":[{"commit":"8967398f82bbe5f2c2cc153d55010bc3025968a4","subject":"db: Resolve additional migration-model mismatches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8967398f82bbe5f2c2cc153d55010bc3025968a4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-10-07 16:40:13.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-10-12 15:57:34.000000000","tz":60},"subject":"db: Add tests to ensure we keep migrations in sync","message":"db: Add tests to ensure we keep migrations in sync\n\nOur database migrations were previously out of sync with the models\nused. These issues are now resolved, but we want to prevent this\nhappening again. Add tests to do just that. oslo.db does most of the\nheavy lifting here and we simply use the tests that this provides.\n\nChange-Id: Iad51e8c4c78f9ed813cb44ad4e13d0d1c84cfc40\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ef7b96b557b235c4d351bf297870eccf9598a639"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ef7b96b557b235c4d351bf297870eccf9598a639"}]},"branch":"refs/heads/master"},"e56e06fd9f03f0bae1d37d5ba801901ae3613beb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-11-04 16:02:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/813223/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/813223/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/813223/3"}}},"commit":{"parents":[{"commit":"613cbb94f9013f7d21f575dd52a9e9d397714539","subject":"db: Resolve additional migration-model mismatches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/613cbb94f9013f7d21f575dd52a9e9d397714539"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-10-07 16:40:13.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-11-04 12:44:29.000000000","tz":0},"subject":"db: Add tests to ensure we keep migrations in sync","message":"db: Add tests to ensure we keep migrations in sync\n\nOur database migrations were previously out of sync with the models\nused. These issues are now resolved, but we want to prevent this\nhappening again. Add tests to do just that. oslo.db does most of the\nheavy lifting here and we simply use the tests that this provides.\n\nChange-Id: Iad51e8c4c78f9ed813cb44ad4e13d0d1c84cfc40\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e56e06fd9f03f0bae1d37d5ba801901ae3613beb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e56e06fd9f03f0bae1d37d5ba801901ae3613beb"}]},"branch":"refs/heads/master"},"4b6d22ba00377c0a12568b65532091e9f4fb1c0d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-12-15 10:47:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/813223/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/813223/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/813223/4"}}},"commit":{"parents":[{"commit":"24137d8029ed1001164a4888fec9f176b2a6bfc6","subject":"db: Resolve additional migration-model mismatches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24137d8029ed1001164a4888fec9f176b2a6bfc6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-10-07 16:40:13.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-12-15 10:01:17.000000000","tz":0},"subject":"db: Add tests to ensure we keep migrations in sync","message":"db: Add tests to ensure we keep migrations in sync\n\nOur database migrations were previously out of sync with the models\nused. These issues are now resolved, but we want to prevent this\nhappening again. Add tests to do just that. oslo.db does most of the\nheavy lifting here and we simply use the tests that this provides.\n\nChange-Id: Iad51e8c4c78f9ed813cb44ad4e13d0d1c84cfc40\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4b6d22ba00377c0a12568b65532091e9f4fb1c0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4b6d22ba00377c0a12568b65532091e9f4fb1c0d"}]},"branch":"refs/heads/master"},"c46ff1d87b598508c290f12886057059e031069e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-01-31 18:17:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/813223/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/813223/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/813223/5"}}},"commit":{"parents":[{"commit":"865ef6b01113d4442111a932dd33b5bc6c873a01","subject":"db: Resolve additional migration-model mismatches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/865ef6b01113d4442111a932dd33b5bc6c873a01"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-10-07 16:40:13.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-31 18:14:26.000000000","tz":0},"subject":"db: Add tests to ensure we keep migrations in sync","message":"db: Add tests to ensure we keep migrations in sync\n\nOur database migrations were previously out of sync with the models\nused. These issues are now resolved, but we want to prevent this\nhappening again. Add tests to do just that. oslo.db does most of the\nheavy lifting here and we simply use the tests that this provides.\n\nChange-Id: Iad51e8c4c78f9ed813cb44ad4e13d0d1c84cfc40\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c46ff1d87b598508c290f12886057059e031069e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c46ff1d87b598508c290f12886057059e031069e"}]},"branch":"refs/heads/master"},"8b5705156ca22a64b65b61d17cd64ff39c7aeb91":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-02-20 19:58:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/813223/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/23/813223/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/23/813223/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/23/813223/6"}}},"commit":{"parents":[{"commit":"c3e8f0a8a285aae2620227253eb87b3743c37352","subject":"db: Resolve additional migration-model mismatches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c3e8f0a8a285aae2620227253eb87b3743c37352"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-10-07 16:40:13.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-02-20 19:05:46.000000000","tz":0},"subject":"db: Add tests to ensure we keep migrations in sync","message":"db: Add tests to ensure we keep migrations in sync\n\nOur database migrations were previously out of sync with the models\nused. These issues are now resolved, but we want to prevent this\nhappening again. Add tests to do just that. oslo.db does most of the\nheavy lifting here and we simply use the tests that this provides.\n\nChange-Id: Iad51e8c4c78f9ed813cb44ad4e13d0d1c84cfc40\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b5705156ca22a64b65b61d17cd64ff39c7aeb91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b5705156ca22a64b65b61d17cd64ff39c7aeb91"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
