)]}'
{"id":"openstack%2Fcinder~813224","triplet_id":"openstack%2Fcinder~master~I4697c4ce53eb8d67d2f7593085e3eb88f239d1dd","project":"openstack/cinder","branch":"master","topic":"bp/remove-sqlalchemy-migrate","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-03-08 00:37:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4697c4ce53eb8d67d2f7593085e3eb88f239d1dd","subject":"docs: Add whereto for testing redirect rules","status":"MERGED","created":"2021-10-08 16:38:33.000000000","updated":"2022-03-08 00:39:49.000000000","submitted":"2022-03-08 00:37:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813224-bp/remove-sqlalchemy-migrate","meta_rev_id":"1f2a790f059bb7e9479fe10315633522f5ef981c","_number":813224,"virtual_id_number":813224,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-08 00:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-03-08 00:19:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-03-08 00:39:49.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":2,"date":"2022-03-08 00:39:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2022-02-21 10:57:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2022-03-08 00:39:49.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-21 10:04:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2022-02-21 11:12:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-03-08 00:19:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-03-08 00:19:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-08 17:27:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-08 21:06:51.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-08 22:56:43.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-10-09 02:14:50.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-09 07:00:20.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-10 07:18:29.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-10-12 18:11:04.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-11-04 16:20:53.000000000","updated_by":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"reviewer":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"state":"CC"},{"updated":"2021-11-04 17:38:51.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-04 18:07:04.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-11-04 21:31:20.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-04 21:43:46.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-11-04 23:26:22.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-11-05 00:43:46.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-05 20:52:01.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-06 17:52:52.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-11-06 20:46:31.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-11-09 19:05:16.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-12-15 12:41:52.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2021-12-15 14:57:33.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2021-12-15 19:28:16.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-16 07:01:09.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-17 15:08:08.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2022-01-10 09:23:58.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2022-01-24 23:59:16.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-01-31 19:54:12.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-02-01 00:13:43.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-01 04:57:13.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-02-01 08:46:17.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-02-01 10:21:56.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-02-02 09:43:33.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2022-02-21 10:04:32.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-02-21 10:57:49.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-02-21 11:12:47.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"1c43820f692dee0cf8f715cb639928e67fca6d70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-10-08 16:38:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5966fd2a0699f804f0337d3bb81743a105351d6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 17:27:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c1441e77c078417aa0d271d89f428ff2 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ab35753a1d840649a6e344578b49e0c : SUCCESS in 9m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d9efd03fc80416b9c577c554d0ee488 : FAILURE in 10m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a58bba24e0e74e4596b5d19e3adc96eb : FAILURE in 10m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f44ed67ed9b4bc885d5fa61fbf68b7c : FAILURE in 17m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/544aebb9a547450d8eb87f75fd46e0e6 : SUCCESS in 13m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8bd3948a1c194a0bb7868e71cad7aaa9 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b62b725c90047c5b71cb28d3a336ae1 : SUCCESS in 7m 57s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4d38375164d74664952371b509f2e9f9 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/05b7d5b69f104f8fbedb63906be11968 : SUCCESS in 7m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a70f98b006d844f18730297481118df2 : SUCCESS in 10m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a5a8ecaca928411eb77272f508164085 : SUCCESS in 44m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cee5071a8e4845ab920b4fd9f0091598 : FAILURE in 6m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f40a0f97fa69c24083cb2961ed1bd4e7653cc7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 17:45:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7302c4439f0e47db99d9cb4ad6ac2b6e : FAILURE in 26m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/45db643c836440a993a33d947e6f2dba : FAILURE in 29m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9702510104ffdbaf857f3301fc1cb1e890cffdb6","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-10-08 21:06:51.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/24/813224/1/check/DellEMC_XtremIO_FC/a7601f1/DellEMC_XtremIO_FC/13 : FAILURE in 35m 19s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/24/813224/1/check/DellEMC_XtremIO_iSCSI/013b3a9/DellEMC_XtremIO_iSCSI/13 : FAILURE in 32m 40s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"70a66d6bd34fe435e31481f8cc1812f14ec75e7f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-08 22:56:43.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/24/813224/1/check/DellEMC_Unity_FC/a2a4944/DellEMC_Unity_FC/13 : FAILURE in 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/24/813224/1/check/DellEMC_Unity_iSCSI/55b1c8e/DellEMC_Unity_iSCSI/12 : FAILURE in 26m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"7bbea4e378f98c6561b216f28378b577081b51c2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-09 02:14:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/24/813224/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/73fca72/73fca72.html : TIMED_OUT in 3h 00m 45s","accounts_in_message":[],"_revision_number":1},{"id":"21fa9bcd287d91578cfb3b5981fd3f0c431ff993","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-09 07:00:20.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-813224-1/ns5_nfs : FAILURE in 59m 14s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-813224-1/ns5_iscsi : FAILURE in 1h 36m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"bc5075c971777415d4d01b9ef3dab05ebb2a4a6b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-09 18:18:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/24/813224/1/check/iscsi-driver-nimble/17a5104/17a5104.html : TIMED_OUT in 3h 00m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0020ade7ce02fd243ef8248d088e3a2f40c34c99","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-10-10 07:18:29.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/24/813224/1/check/dsvm-tempest-cisco-zonemanager-job/2046a37/ : FAILURE in 2h 59m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f7c4d51404625eb70a0861321618b01920899087","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-10-12 16:01:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4c55eadce9123073ad06d4e528182d67e35eab18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 16:32:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ccdcc9a34e34d83b1e316d0d3bbe8b2 : FAILURE in 18m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82fb7c2a527e4ebabac11b1ea335d72f : FAILURE in 23m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d48e832c1efa0596f9c1bfcabb2242c196d04206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 16:44:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9f48eb667c044fc69d1787a263a34567 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51772dfa1c4d4db7ab447524cf0e4ffe : SUCCESS in 10m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7372c557c584e25819eb40a9350687a : FAILURE in 14m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c76c1d53861649db93bdd6e9af8bcac5 : FAILURE in 11m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d73f202276e04d71b225e0d9c6f101cd : FAILURE in 12m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e350dc57d6394696aec24a6215b42614 : SUCCESS in 13m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9673e2034ad24d399667446232794157 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87a68c69acf94b16aa05c58a6a082c28 : SUCCESS in 8m 08s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9c74079a75aa4dd9aa8f30865be737f6 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d36fc4058c964cbf89eede4bab230ba5 : SUCCESS in 7m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2549f28144974d219e9ebd5fd391e517 : SUCCESS in 8m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/59e32361ace24988893c93f0a0aabcdc : SUCCESS in 34m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6f9d742892df466a9c5b847b1c875693 : FAILURE in 6m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d0c260587ca89a1651a43edd1c29383bb724ccb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-10-12 18:11:04.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813224/2/iscsi/ : FAILURE in 1h 07m 49s\n- cinder-smb http://cloudbase-ci.com/cinder/813224/2/smb/ : SUCCESS in 48m 01s","accounts_in_message":[],"_revision_number":2},{"id":"21556c97177bec7a2254e395ff5ac192fdfaad31","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-10-12 18:24:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/24/813224/2/check/dsvm-tempest-cisco-zonemanager-job/7896eca/ : FAILURE in 29m 39s","accounts_in_message":[],"_revision_number":2},{"id":"8fa614f1d9518b6bdb8196c09a783e19ac1dca33","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-12 18:53:30.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-813224-2/ns5_nfs : FAILURE in 33m 22s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-813224-2/ns5_iscsi : FAILURE in 35m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"669291d1955cf239e8b2ed7e3bf346088032242f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-12 19:13:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/24/813224/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/452d645/452d645.html : FAILURE in 1h 18m 08s","accounts_in_message":[],"_revision_number":2},{"id":"c60decdc10f73286817f4e95c6a58360aa4c1aa8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-12 19:23:57.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/24/813224/2/check/iscsi-driver-primera/100649f/100649f.html : FAILURE in 1h 28m 15s","accounts_in_message":[],"_revision_number":2},{"id":"39b32d8f2c5f974fdfe5925b477d4c0b79368d4a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-12 21:22:55.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/24/813224/2/check/iscsi-driver-alletra-6k/e454f23/e454f23.html : SUCCESS in 1h 39m 52s","accounts_in_message":[],"_revision_number":2},{"id":"93eb931235b8f06e6161a62d08f169d09d45922e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-12 22:53:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/24/813224/2/check/iscsi-driver-nimble/2f2b02e/2f2b02e.html : SUCCESS in 1h 48m 39s","accounts_in_message":[],"_revision_number":2},{"id":"1919fd428da60111aa9f7b9c60f2676ede3a59e6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-13 01:03:18.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/24/813224/2/check/DellEMC_Unity_FC/5106639/DellEMC_Unity_FC/32 : FAILURE in 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/24/813224/2/check/DellEMC_Unity_iSCSI/edd13e4/DellEMC_Unity_iSCSI/31 : FAILURE in 2h 48m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"4664b0b800c8e043d3904fb72c8824b6dd768ebe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-04 16:02:06.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"dd5f90256af4d32516cd9a8d037271044398b8fa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2021-11-04 16:20:53.000000000","message":"Patch Set 3:\n\nBuild Failed. ibm-gpfs-dsvm-tempest-full: s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/101/index.html","accounts_in_message":[],"_revision_number":3},{"id":"f8d1e738a80086ca48fdd417e357ec832ab17020","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-04 16:31:43.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813224/3/iscsi/ : FAILURE in 14m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/813224/3/smb/ : FAILURE in 13m 59s","accounts_in_message":[],"_revision_number":3},{"id":"3237db3f695021f898479940eefdc22d134ca5c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 16:40:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cdf85fd9f79b4d2ba654ea89ea3e7eba : SUCCESS in 18m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7faebd1ccab84878847b2f5f11b805d6 : SUCCESS in 23m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b275008af4312e06761879a619c85b1916c5a7c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 16:56:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/00ff5b3350ee44f1be7031965ebd40f1 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90452c3db5f44fa2b0128754ad02931c : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7cf57f76cee4cac845f6283e23df2ce : SUCCESS in 11m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63e6df7adc674a4db0b0257d0ac1f56a : SUCCESS in 12m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ce0e4428ae9497187da024f54ddf305 : SUCCESS in 14m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c693a39a22846b693e834be5ce1d2a4 : SUCCESS in 15m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/359d5a66f4c949139f245655eef0dd41 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc9f0975de0c4bf3bd018c0dd51bdb5e : SUCCESS in 8m 09s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b7b9ead15f994f3b99b8c850b2d23953 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9917e2a7926545ada113ac2656a70cb8 : SUCCESS in 12m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c4492b8672754eb1981e861304bf5343 : SUCCESS in 8m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/613c72b50b024a8a84e513a2f11ff40c : FAILURE in 27m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4e7a4025efe04505bf08cf1e1b65d4c9 : FAILURE in 7m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e027e442473a12a3d6a39ee5f038d424707e4e4e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-04 17:38:51.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/24/813224/3/upstream-check/cinder-cDOT-iSCSI/fc0b3c0/ : SUCCESS in 38m 21s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/24/813224/3/upstream-check/cinder-cDOT-NFS/6e80f25/ : SUCCESS in 40m 42s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/24/813224/3/upstream-check/cinder-cDOT-FCP/8670925/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"bfa28dd979316135d18c924a2e5193431e7a9174","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-11-04 18:07:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/24/813224/3/upstream-solidfire-check/cinder-solidfire-iSCSI/0125574/ : SUCCESS in 1h 07m 51s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"ff9376bf2eeb59e6b2e0262fcef067e70a0449b0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-04 19:44:07.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/24/813224/3/check/iscsi-driver-primera/954a1ae/954a1ae.html : FAILURE in 1h 20m 49s","accounts_in_message":[],"_revision_number":3},{"id":"ff168a2508ae870f63f77df0c3187efad97b511d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-11-04 21:31:20.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/24/813224/3/check/DellEMC_PowerFlex/2a77c62/DellEMC_PowerFlex/72 : FAILURE in 1h 52m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"0cb08f7433ba69b7d499917ec512e8bfa93854b9","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-11-04 21:43:46.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/24/813224/3/check/DellEMC_PowerStore_iSCSI/f5243e2/DellEMC_PowerStore_iSCSI/33 : SUCCESS in 1h 45m 23s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/24/813224/3/check/DellEMC_PowerStore_FC/f8f4f05/DellEMC_PowerStore_FC/40 : SUCCESS in 1h 17m 35s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"108c39a8c4b6ad5f6036f60b852401cfd2ecb400","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-04 22:53:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/24/813224/3/check/iscsi-driver-alletra-6k/81d84d7/81d84d7.html : SUCCESS in 1h 39m 44s","accounts_in_message":[],"_revision_number":3},{"id":"16d847164a70e4ceaf733ef269913013ae5ef050","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-11-04 23:26:22.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI http://os-logs.tristero.net/dell-ci/813224-3/PVME-iSCSI : SUCCESS in 59:43.22\n- PVME-FC http://os-logs.tristero.net/dell-ci/813224-3/PVME-FC : SUCCESS in 58:40.07","accounts_in_message":[],"_revision_number":3},{"id":"bc350ac2a0fb25d89ca9d40ecd92011ce4371823","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-04 23:52:10.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/24/813224/3/check/iscsi-driver-nimble/a3b5f34/a3b5f34.html : FAILURE in 1h 13m 16s","accounts_in_message":[],"_revision_number":3},{"id":"92bcbab9c61036884d8b65042eef6d58b0120ea1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-05 00:24:50.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/24/813224/3/check/dsvm-tempest-cisco-zonemanager-job/125236d/ : SUCCESS in 1h 09m 21s","accounts_in_message":[],"_revision_number":3},{"id":"59659fe10b1ffbb7e971dd9899e5c68cbe5ecfe2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-11-05 00:43:46.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/24/813224/3/check/DellEMC_VNX_iSCSI/66e307d/DellEMC_VNX_iSCSI/69 : FAILURE in 2h 55m 37s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/24/813224/3/check/DellEMC_VNX_FC/e39ac72/DellEMC_VNX_FC/68 : SUCCESS in 2h 08m 50s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"4d17da409b017c6ba1f51cd61dd1dfb9de9c1595","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-05 03:01:39.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-813224-3/ns5_nfs : SUCCESS in 4h 53m 51s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-813224-3/ns5_iscsi : SUCCESS in 4h 38m 49s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"057c14caca6c03878e95a5bf506dbd95e8b9e415","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-05 20:52:01.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/813224/3/20211106024708/NEC-ISCSI : SUCCESS in 2h 50m 12s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/813224/3/20211106024708/NEC-FC : SUCCESS in 3h 04m 50s","accounts_in_message":[],"_revision_number":3},{"id":"98998eaaba92b32d8303bd556fd9e32aa505802d","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-11-06 17:52:52.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/813224-3/HPMSA-iSCSI : SUCCESS in 1:08:54\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/813224-3/HPMSA-FC : SUCCESS in 44:15.76","accounts_in_message":[],"_revision_number":3},{"id":"3b332243152a1de0842457eee48dd0766a0325f1","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-11-06 20:46:31.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/813224-3/Lenovo-iSCSI : SUCCESS in 1:26:48\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/813224-3/Lenovo-FC : SUCCESS in 58:35.08","accounts_in_message":[],"_revision_number":3},{"id":"d726500e273df27b698535f4d25ff76a97193a02","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-11-09 19:05:16.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/24/813224/3/check/dsvm-tempest-TOYOU-ACS5000-driver/c907891 : FAILURE in 2h 17m 42s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/24/813224/3/check/fc-test/85ad3fc : FAILURE in 2m 26s","accounts_in_message":[],"_revision_number":3},{"id":"275bc521b9ae5b8618b93060a69f1b2676c4339e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-15 10:47:01.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"a55daa8de3948b3dfae9f1b4956ce6348774a44d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 11:25:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49303145a6594a0f92fd9bd231299f0b : SUCCESS in 18m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ef1a759847af4abf82fdd564008227e4 : SUCCESS in 21m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"30ac9b3b3177b51e643b3397bf9fd3467873f9fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 11:36:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/16a220213e314ed7a62179178df46021 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/415afa15117d42d195750f3928f9a627 : SUCCESS in 8m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/467fb54df11b4b6b9feec1c69595d64c : SUCCESS in 13m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c80adf54ca8a4e65b09f3d9cb3140457 : SUCCESS in 12m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee9a5960aea84c0d8161c3251ab6d686 : SUCCESS in 15m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b9098359179a4bcab0e931b3c5828039 : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1d5bdaf1dea4121ad257a3f179549db : SUCCESS in 8m 56s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b0376415e2674af08701c1a2a723a6a3 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2622e72c9ab849a4877d92bf123b3464 : SUCCESS in 11m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a47b5f65b8e041f78187cec078ab317e : SUCCESS in 8m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1a406cb3dfc944e0b8947813a0e045db : SUCCESS in 41m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/60ace57c61de4e6f8ed13c5fd4ec6b30 : SUCCESS in 8m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e2a59d89598aa9f268f00f60ac7b836c0683dd38","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-12-15 11:44:26.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"7dfb886fe5c8dc09a023945d665199236676e083","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-12-15 11:48:07.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/24/813224/4/upstream-solidfire-check/cinder-solidfire-iSCSI/a691eca/ : FAILURE in 7m 01s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"dfd46504bddd7e4247fd1efeb1356dd2bbd3852b","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-12-15 12:41:52.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-24-813224-4/1639568212 : SUCCESS in 31m 55s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"4aa1dcc60ed59685057b6bc0d43bfc1d6f418339","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-15 13:22:50.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813224/4/iscsi/ : FAILURE in 1h 31m 00s\n- cinder-smb http://cloudbase-ci.com/cinder/813224/4/smb/ : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":4},{"id":"594b0bc3c2d7a7aa35ffc247a104f3916718f6ee","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-15 14:57:33.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/24/813224/4/check/iscsi-driver-alletra-6k/889af32/889af32.html : FAILURE in 1h 02m 55s","accounts_in_message":[],"_revision_number":4},{"id":"d69b707c6527ed727bb9b8d0ef2bd5e26579f0be","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-15 19:28:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ab9da38f420b4ea38d84683df4d1f45b : SUCCESS in 1h 24m 18s","accounts_in_message":[],"_revision_number":4},{"id":"426da0292297948b01998c11a56b07163df9792e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-15 19:35:50.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-813224-4/ns5_nfs : SUCCESS in 4h 36m 27s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-813224-4/ns5_iscsi : SUCCESS in 4h 43m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"713a264455f2b5321d47302886a02481e482ba42","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-15 19:56:23.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/24/813224/4/check/iscsi-driver-nimble/40e9a80/40e9a80.html : SUCCESS in 1h 50m 27s","accounts_in_message":[],"_revision_number":4},{"id":"8be7fe60c24bef39a0bd570614df259e80193986","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-12-16 07:01:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/24/813224/4/ : SUCCESS in 1h 26m 48s","accounts_in_message":[],"_revision_number":4},{"id":"462a5df17d284c0c2a26a1aefdb99322a487ab69","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-12-17 03:31:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/813224/4/20211217062830/NEC-FC : SUCCESS in 2h 41m 21s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/813224/4/20211217062830/NEC-ISCSI : SUCCESS in 6h 02m 32s","accounts_in_message":[],"_revision_number":4},{"id":"9283b5486791aaf2635fb49979049731eb03074b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-12-17 15:08:08.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ee067a4a7f310faf668e3d7e6dec496ec6eef654","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-12-17 15:10:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2dc39570108d666ae261043774429a35c675c2b0","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-01-10 09:23:58.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"a523b2a5b20b86589d427500386b5b0d48a6a9bb","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-01-24 23:59:16.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3e915565329a03b87b1094efd1c9791f80b3c0d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-31 18:17:06.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"0b1f2c487867f8092e035109a68c94c1c4bd9a14","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-01-31 18:21:34.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"928d2c6078db857c7a177e0ebeb0dbac0059f795","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-31 19:28:48.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813224/5/iscsi/ : SUCCESS in 1h 03m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/813224/5/smb/ : FAILURE in 1h 03m 55s","accounts_in_message":[],"_revision_number":5},{"id":"77d84658b55f5cdf845da211ba94f66571a4a83c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 19:46:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5aa490863088441c8016114c062fb9c4 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bd9c7ff0fb2453b91c8f532385d7992 : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b6e9a2a1e224832b2a794af544d2bde : SUCCESS in 10m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb7f0f455c1447e382dcad9559544e3b : SUCCESS in 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab03642495314512908481ecf1f238d0 : SUCCESS in 14m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9e02ea85856f40fda4dbfa0e00bd2420 : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/096ba0274eba48c7b6d8133d7daaf44a : SUCCESS in 8m 48s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e1cb6b2bf34644f1854d5af53c72dd2f : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/339a7bf695e04d9db77761573ea5c4f7 : SUCCESS in 8m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7966a8b361334dadbacba6e937a0aed9 : SUCCESS in 8m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/86383b51a38c4402ae3563aae6567a42 : SUCCESS in 50m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b59f8788a3d64589afcc54dc08fbf1b4 : SUCCESS in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"68a6e7a89a9c6f80a290f4a5eadfbe150d037ff7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-31 19:54:12.000000000","message":"Patch Set 5:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/31_01_2022_19:46:50\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"d77494affb0b203ce9cb140d718e04181a4adee2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-31 19:56:45.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/24/813224/5/check/DellEMC_PowerFlex/c502136/DellEMC_PowerFlex/340 : FAILURE in 9m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"fa628c61a992e4669bcff14a594c086856fa45cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 20:02:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8d2fafafc35486fbc9d69f250e7984c : FAILURE in 33m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/48e35acf22144660836ee830568aef8e : FAILURE in 17m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"67620266cf8e54fb3ce5811042564cf5b3fa9bef","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-31 20:03:55.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/24/813224/5/check/DellEMC_PowerFlex/4ec2fb7/DellEMC_PowerFlex/342 : FAILURE in 1m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"264bba19c3c8799ff4da72a99281f75b6875e206","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-01-31 20:58:38.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/24/813224/5/upstream-solidfire-check/cinder-solidfire-iSCSI/c2c45c8/ : SUCCESS in 1h 06m 17s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"64a0b76f786e35134f44fa769ff855abbd2af7e9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-01-31 21:03:21.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/24/813224/5/check/iscsi-driver-primera/ec7399d/ec7399d.html : SUCCESS in 1h 18m 17s","accounts_in_message":[],"_revision_number":5},{"id":"4188205727c7fe4ceaf244a2440141fa79f6f38d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-31 21:31:06.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/24/813224/5/upstream-check/cinder-cDOT-iSCSI/d960424/ : SUCCESS in 38m 19s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/24/813224/5/upstream-check/cinder-cDOT-NFS/24932a4/ : SUCCESS in 37m 04s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/24/813224/5/upstream-check/cinder-cDOT-FCP/cc73b55/ : FAILURE in 14m 15s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"f4892b3ba54f98620ce84a0f4a395d1ab9a96227","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-01-31 21:38:05.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/24/813224/5/check/iscsi-driver-alletra-6k/5911869/5911869.html : SUCCESS in 1h 37m 28s","accounts_in_message":[],"_revision_number":5},{"id":"35854fc82298488c7c0321c7f999a39022da32c3","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-01-31 21:48:51.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-24-813224-5/1643658404 : SUCCESS in 27m 59s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":5},{"id":"d85dbdd3e0986abefb5a3cf4b248e470526c8ca8","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-01-31 22:49:17.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-24-813224-5/1643659327 : SUCCESS in 26m 13s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":5},{"id":"83f2fc4e8c5b75f0584e89cb177088b4e1eeebb2","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-02-01 00:13:43.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/24/813224/5/check/DellEMC_PowerMAX_FC/1eb0c6e/DellEMC_PowerMAX_FC/254 : FAILURE in 52m 39s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/24/813224/5/check/DellEMC_PowerMAX_iSCSI/3f99ca9/DellEMC_PowerMAX_iSCSI/267 : FAILURE in 52m 49s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"bf0590ca651a6e647e3a2e8939d97e717ff3fa9a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-02-01 00:45:32.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/24/813224/5/check/DellEMC_PowerStore_iSCSI/45a25e4/DellEMC_PowerStore_iSCSI/293 : SUCCESS in 1h 21m 58s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/24/813224/5/check/DellEMC_PowerStore_FC/6c17523/DellEMC_PowerStore_FC/310 : SUCCESS in 1h 06m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"9855f74dd7dea5f0729a084a850bdf6a16f43f4f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-01 01:08:45.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/24/813224/5/check/dsvm-tempest-cisco-zonemanager-job/c1034ce/ : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":5},{"id":"0710e3da9e1112bde8ebbd88f888d56a711b721a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-01 02:04:18.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-813224-5/ns5_nfs : SUCCESS in 4h 42m 48s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-813224-5/ns5_iscsi : SUCCESS in 4h 46m 33s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"6e709b370595da5d2f0e47100b9d9cfd5fca5183","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-01 02:48:10.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/24/813224/5/check/iscsi-driver-nimble/6a05492/6a05492.html : SUCCESS in 1h 50m 08s","accounts_in_message":[],"_revision_number":5},{"id":"9d4a2265a76ca1a5b2d8735bad4eddf273d678b1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-02-01 03:05:47.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/24/813224/5/check/DellEMC_Unity_FC/0435989/DellEMC_Unity_FC/493 : FAILURE in 1h 23m 52s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/24/813224/5/check/DellEMC_Unity_iSCSI/ed29ed5/DellEMC_Unity_iSCSI/487 : FAILURE in 4m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"3c0262214e4e255f2f60ffcdf6670535867f092b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-02-01 04:57:13.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/24/813224/5/check/DellEMC_SC_FC/711b17d/DellEMC_SC_FC/277 : FAILURE in 1h 08m 28s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/24/813224/5/check/DellEMC_SC_iSCSI/4c755c3/DellEMC_SC_iSCSI/273 : SUCCESS in 1h 45m 07s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"0467e307ff951c873df99c914aa1ea4285ffc5d5","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-01 06:00:32.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/24/813224/5/check/DellEMC_XtremIO_FC/2eacd4c/DellEMC_XtremIO_FC/607 : SUCCESS in 57m 06s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/24/813224/5/check/DellEMC_XtremIO_iSCSI/264c9ce/DellEMC_XtremIO_iSCSI/617 : SUCCESS in 1h 27m 12s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"51abace065dbcfd2a7c5b81854fac0a4fb059b58","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-01 06:42:03.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/813224-5/HPMSA-iSCSI/ : SUCCESS in 50:26.38\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/813224-5/HPMSA-FC/ : SUCCESS in 44:01.30","accounts_in_message":[],"_revision_number":5},{"id":"0cc644f43ad33c4529512c6b5f0ab9070bcf00e7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-02-01 08:46:17.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/813224/5/20220201151456/NEC-V-FC : SUCCESS in 2h 28m 01s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/813224/5/20220201151456/NEC-V-ISCSI : SUCCESS in 2h 31m 18s","accounts_in_message":[],"_revision_number":5},{"id":"fdba00fbeadea07d589d00657c99903124d182df","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-01 09:09:05.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/813224-5/Lenovo-iSCSI/ : SUCCESS in 1:02:12\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/813224-5/Lenovo-FC/ : SUCCESS in 56:35.82","accounts_in_message":[],"_revision_number":5},{"id":"d77828cdc95d322ddd9c4ca491baa873b970ebeb","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-01 09:49:25.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/813224-5/PVME-iSCSI/ : SUCCESS in 1:01:32\n- PVME-FC https://os-logs.seagate.com/dell-ci/813224-5/PVME-FC/ : SUCCESS in 1:02:33","accounts_in_message":[],"_revision_number":5},{"id":"df00ebeb9f322fe95d30658d0b7ea029009483ba","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-02-01 10:21:56.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/813224-5/Seagate-iSCSI/ : SUCCESS in 1:06:22\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/813224-5/Seagate-FC/ : SUCCESS in 1:01:42","accounts_in_message":[],"_revision_number":5},{"id":"d990bfee1d29624a84978992bd4edb1628e461c4","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-02 04:43:28.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/24/813224/5/ : SUCCESS in 1h 47m 52s","accounts_in_message":[],"_revision_number":5},{"id":"8e9c691a842dfe29d36a79231cd4d1b20a931de4","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2022-02-02 09:43:33.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e183fea9e6089a24da17552ba52693cf7660712e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-02-04 04:57:34.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/813224/5/20220204103700/NEC-ISCSI : SUCCESS in 3h 19m 51s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/813224/5/20220204103700/NEC-FC : SUCCESS in 3h 20m 30s","accounts_in_message":[],"_revision_number":5},{"id":"b1d3f66512d54af966e4df4e007957ba7533c04b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2022-02-18 14:13:56.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"72c8812240df2d7c09075694a37dc52afd61191c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-20 19:58:21.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"5a0f7b6795d9ef101257436bece83076172605bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 20:39:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1e0f1d9b6754292b36c88f746b1d52f : FAILURE in 31m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9bce7f3dc474f699653304f9949ca9f : FAILURE in 33m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0ed4210e6295bfa99e93254f0d69ef64bc6f30e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 21:22:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7101920d6d1d49a899bfb4d9d39323c3 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee5f3d0db8994a259b2fa9514611800c : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce87cd4e17fa4ad1a1e1c73c51976e79 : SUCCESS in 12m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/942fe1044e024baa9cfe5d4d47b4e562 : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c90dcb6edc724f32b5a6e2c54b6e133f : SUCCESS in 12m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f5f3319f875145b8bfabd88801757dfc : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1c68e2279f240f4ad77aca8d7903b2c : SUCCESS in 7m 16s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a9f6becda6ed4b789edafcc2b2db2631 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b88c8473a6a84753a514c2ce50a0ed18 : SUCCESS in 10m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bbff4a4139c345b9a34d83bce140f437 : SUCCESS in 8m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4e61b73b6ad3426fa1685ae2eab0f173 : SUCCESS in 1h 16m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/930f3bcefbef4f93b864c6907a9f13e2 : SUCCESS in 5m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dc4d1680fc8e6a2fc1bf8f44c4f3401a5d40dd13","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-20 22:34:06.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-24-813224-6/1645392133 : SUCCESS in 17m 30s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":6},{"id":"61baf6001cd1ec498fc97d9762fbc5c3b30eba89","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-21 00:58:58.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813224/6/iscsi/ : SUCCESS in 1h 09m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/813224/6/smb/ : SUCCESS in 58m 12s","accounts_in_message":[],"_revision_number":6},{"id":"b654ade3b27ae57fd9d03f899370fd467a877034","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-21 07:17:50.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/24/813224/6/check/dsvm-tempest-cisco-zonemanager-job/6450552/ : SUCCESS in 1h 08m 04s","accounts_in_message":[],"_revision_number":6},{"id":"4a97de685e8bac09397220b9a5de6d0b763d76db","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-21 10:03:19.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/24/813224/6/check/iscsi-driver-alletra-6k/041d994/041d994.html : SUCCESS in 1h 38m 15s","accounts_in_message":[],"_revision_number":6},{"id":"ba900a3b116a476ee9043e63a56837702b69e07b","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-02-21 10:04:32.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"f779f106ee73025625d294f48cf23a6f2ca0b24a","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-02-21 10:57:49.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"19d6fcb7e528fcdf56d3adc36cd63a797b5afd93","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-02-21 11:12:47.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"504f916ac38772a1ccff90a14a3dee55e332d85d","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-21 12:12:42.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/813224-6/HPMSA-iSCSI/ : SUCCESS in 48:52.32\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/813224-6/HPMSA-FC/ : SUCCESS in 44:22.84","accounts_in_message":[],"_revision_number":6},{"id":"e898db532779825c7cc19054057a3895fad4775c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-21 14:39:57.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-813224-6/ns5_nfs : SUCCESS in 5h 2m 4s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-813224-6/ns5_iscsi : SUCCESS in 5h 9m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"0b9088f0eff0d305089ac935a47d28f3cd8cc262","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-21 15:31:11.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/813224-6/Lenovo-iSCSI/ : SUCCESS in 1:03:06\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/813224-6/Lenovo-FC/ : SUCCESS in 57:18.12","accounts_in_message":[],"_revision_number":6},{"id":"7729c76472331fd01c6faca56ddfa451588e9e3d","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-21 16:01:06.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/813224-6/PVME-iSCSI/ : SUCCESS in 1:03:45\n- PVME-FC https://os-logs.seagate.com/dell-ci/813224-6/PVME-FC/ : SUCCESS in 1:02:34","accounts_in_message":[],"_revision_number":6},{"id":"518ad408dfab3c3b57e4691b83ab24b30ca2909d","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-02-21 16:33:14.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/813224-6/Seagate-iSCSI/ : SUCCESS in 1:03:31\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/813224-6/Seagate-FC/ : SUCCESS in 1:03:03","accounts_in_message":[],"_revision_number":6},{"id":"602f802f729c4fdc18e710859dbfbae5b6462260","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-02-21 23:59:40.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/24/813224/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/4cc4459/4cc4459.html : FAILURE in 2h 02m 17s","accounts_in_message":[],"_revision_number":6},{"id":"3e6452a764661a033ea666402bd587bbe0783196","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-22 00:37:00.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/24/813224/6/ : SUCCESS in 1h 42m 38s","accounts_in_message":[],"_revision_number":6},{"id":"c6eeef571a6782fe2318748d7755f1af532e6d87","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-02-24 05:22:40.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/813224/6/20220224110048/NEC-FC : SUCCESS in 3h 15m 20s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/813224/6/20220224110048/NEC-ISCSI : SUCCESS in 3h 21m 49s","accounts_in_message":[],"_revision_number":6},{"id":"232e0adc8c12306eed2e95186d87648f56c6bf71","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-02-25 16:12:31.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/813224/6/20220225222325/NEC-V-FC : SUCCESS in 2h 28m 40s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/813224/6/20220225222325/NEC-V-ISCSI : SUCCESS in 2h 49m 03s","accounts_in_message":[],"_revision_number":6},{"id":"e76db2f8027e183d0addb73bac9564849a56a7f8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-08 00:19:18.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"54f25abe554f1a6cc22482368305c00ea7051be1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 00:20:05.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b3a32f1977fce03beddc8a1f1dee910d46de5434","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 00:37:57.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/77ff433fda124b1880379da2b7bd4754 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42b7faa66aa5493c95a6f37e7ff18b55 : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac371fddf16b499cbd470c6f7bb6c3c1 : SUCCESS in 12m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37c6bbf3647b4c3790138a00c9cfd481 : SUCCESS in 13m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63abee7be6b942d881961704674d0795 : SUCCESS in 14m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/181ccb28e6954b06822c1d14e19a9d09 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90b84e42adce4e488fcb516f513026bc : SUCCESS in 8m 59s","accounts_in_message":[],"_revision_number":6},{"id":"454617f5b33efb414fa68cae4a5e74779b1d83b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 00:37:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"85f2e4e43be775a3e0fceb21cf39b5cbfc43aff4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 00:39:49.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/dfcbe948dfe441299636911ac218d4cf : SUCCESS in 1m 02s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d9e140c547749b59abdb361054a2ef6 : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e18b49fe9c0455d9b0b39e8a88e2d64 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"31303ed65e400c6df7d71e03ba030a1bada17118","revisions":{"bf01388e2c1796a0e2839f10548261ab1517c826":{"kind":"REWORK","_number":1,"created":"2021-10-08 16:38:33.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/24/813224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/813224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/813224/1"}}},"commit":{"parents":[{"commit":"cd59eb7c1c660da842b7ec73d984a43170cfaf76","subject":"db: Add tests to ensure we keep migrations in sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd59eb7c1c660da842b7ec73d984a43170cfaf76"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-24 16:31:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-10-08 16:38:26.000000000","tz":60},"subject":"docs: Add whereto for testing redirect rules","message":"docs: Add whereto for testing redirect rules\n\nUse whereto to test that the redirect rules do what we expect. This has\nbeen used in nova-land for years and is helpful as things move around.\n\nChange-Id: I4697c4ce53eb8d67d2f7593085e3eb88f239d1dd\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf01388e2c1796a0e2839f10548261ab1517c826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf01388e2c1796a0e2839f10548261ab1517c826"}]},"branch":"refs/heads/master"},"8166a81cdd250aecf184bd2878098228404c8bf2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-12 16:01:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/24/813224/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/813224/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/813224/2"}}},"commit":{"parents":[{"commit":"ef7b96b557b235c4d351bf297870eccf9598a639","subject":"db: Add tests to ensure we keep migrations in sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ef7b96b557b235c4d351bf297870eccf9598a639"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-24 16:31:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-10-12 15:57:34.000000000","tz":60},"subject":"docs: Add whereto for testing redirect rules","message":"docs: Add whereto for testing redirect rules\n\nUse whereto to test that the redirect rules do what we expect. This has\nbeen used in nova-land for years and is helpful as things move around.\n\nChange-Id: I4697c4ce53eb8d67d2f7593085e3eb88f239d1dd\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8166a81cdd250aecf184bd2878098228404c8bf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8166a81cdd250aecf184bd2878098228404c8bf2"}]},"branch":"refs/heads/master"},"893251ccdd0643a6ab139e326cf655a25b9fb653":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-11-04 16:02:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/24/813224/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/813224/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/813224/3"}}},"commit":{"parents":[{"commit":"e56e06fd9f03f0bae1d37d5ba801901ae3613beb","subject":"db: Add tests to ensure we keep migrations in sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e56e06fd9f03f0bae1d37d5ba801901ae3613beb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-24 16:31:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-11-04 12:44:29.000000000","tz":0},"subject":"docs: Add whereto for testing redirect rules","message":"docs: Add whereto for testing redirect rules\n\nUse whereto to test that the redirect rules do what we expect. This has\nbeen used in nova-land for years and is helpful as things move around.\n\nChange-Id: I4697c4ce53eb8d67d2f7593085e3eb88f239d1dd\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/893251ccdd0643a6ab139e326cf655a25b9fb653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/893251ccdd0643a6ab139e326cf655a25b9fb653"}]},"branch":"refs/heads/master"},"1b07dd3c73e84b854bdaaa4a172db83d329555e8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-12-15 10:47:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/24/813224/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/813224/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/813224/4"}}},"commit":{"parents":[{"commit":"4b6d22ba00377c0a12568b65532091e9f4fb1c0d","subject":"db: Add tests to ensure we keep migrations in sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4b6d22ba00377c0a12568b65532091e9f4fb1c0d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-24 16:31:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-12-15 10:46:52.000000000","tz":0},"subject":"docs: Add whereto for testing redirect rules","message":"docs: Add whereto for testing redirect rules\n\nUse whereto to test that the redirect rules do what we expect. This has\nbeen used in nova-land for years and is helpful as things move around.\n\nChange-Id: I4697c4ce53eb8d67d2f7593085e3eb88f239d1dd\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1b07dd3c73e84b854bdaaa4a172db83d329555e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1b07dd3c73e84b854bdaaa4a172db83d329555e8"}]},"branch":"refs/heads/master"},"238467221f93215c5add898a8e94365c2f338780":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-01-31 18:17:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/24/813224/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/813224/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/813224/5"}}},"commit":{"parents":[{"commit":"c46ff1d87b598508c290f12886057059e031069e","subject":"db: Add tests to ensure we keep migrations in sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c46ff1d87b598508c290f12886057059e031069e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-24 16:31:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-31 18:14:26.000000000","tz":0},"subject":"docs: Add whereto for testing redirect rules","message":"docs: Add whereto for testing redirect rules\n\nUse whereto to test that the redirect rules do what we expect. This has\nbeen used in nova-land for years and is helpful as things move around.\n\nChange-Id: I4697c4ce53eb8d67d2f7593085e3eb88f239d1dd\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/238467221f93215c5add898a8e94365c2f338780"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/238467221f93215c5add898a8e94365c2f338780"}]},"branch":"refs/heads/master"},"31303ed65e400c6df7d71e03ba030a1bada17118":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-02-20 19:58:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/24/813224/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/813224/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/813224/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/813224/6"}}},"commit":{"parents":[{"commit":"8b5705156ca22a64b65b61d17cd64ff39c7aeb91","subject":"db: Add tests to ensure we keep migrations in sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b5705156ca22a64b65b61d17cd64ff39c7aeb91"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-24 16:31:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-02-20 19:05:46.000000000","tz":0},"subject":"docs: Add whereto for testing redirect rules","message":"docs: Add whereto for testing redirect rules\n\nUse whereto to test that the redirect rules do what we expect. This has\nbeen used in nova-land for years and is helpful as things move around.\n\nChange-Id: I4697c4ce53eb8d67d2f7593085e3eb88f239d1dd\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31303ed65e400c6df7d71e03ba030a1bada17118"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31303ed65e400c6df7d71e03ba030a1bada17118"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
