)]}'
{"id":"openstack%2Fcinder~813227","triplet_id":"openstack%2Fcinder~master~Ic1950ff89021a89de397619eef17f8100eb3d847","project":"openstack/cinder","branch":"master","topic":"bp/remove-sqlalchemy-migrate","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-03-08 15:45:44.000000000","reason":"Change was submitted"},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2022-03-03 14:31:28.000000000","reason":"Sofia Enriquez replied on the change"}},"hashtags":[],"change_id":"Ic1950ff89021a89de397619eef17f8100eb3d847","subject":"docs: Remove unnecessary \u0027blockstorage-\u0027 prefix","status":"MERGED","created":"2021-10-08 16:38:33.000000000","updated":"2022-03-08 15:53:18.000000000","submitted":"2022-03-08 15:45:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813227-bp/remove-sqlalchemy-migrate","meta_rev_id":"f107313a4f76bd8949ed8d295ea7724194399cf4","_number":813227,"virtual_id_number":813227,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-08 15:45:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-03-01 04:01:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-03 14:31:28.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-03 14:31:28.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-03-01 04:01:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-08 17:01:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-08 23:01:41.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-09 01:10:55.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-10-09 06:59:21.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-09 08:30:23.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-10 16:35:10.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-10-12 18:43:00.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-10-22 12:56:39.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-04 16:29:46.000000000","updated_by":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"reviewer":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"state":"CC"},{"updated":"2021-11-04 17:52:03.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-04 18:22:01.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-11-05 01:16:55.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-05 01:20:31.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-11-05 02:26:49.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-11-07 08:56:18.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-11-07 11:17:21.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-11-07 15:24:25.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-11-10 00:15:00.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-11-15 01:23:36.000000000","updated_by":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"reviewer":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"state":"CC"},{"updated":"2021-12-15 13:12:56.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2021-12-15 13:56:55.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2021-12-15 17:14:26.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-15 22:21:10.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-31 19:38:53.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-01-31 23:20:57.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-01 03:12:00.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-02-01 20:42:53.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-02-20 21:00:30.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-21 21:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2022-03-01 04:01:07.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-03-02 20:12:47.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"}],"messages":[{"id":"91bafb150c006610397f86b980acf508aab45a68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-10-08 16:38:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f36eb56f02b20d518e2ac46c1b82f8d80c22dbbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 17:01:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a0ff8d44c034ef08d62c3bb621e0a1b : FAILURE in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/504dc475c0dc49388852e4e28aad3836 : SUCCESS in 13m 14s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/817b23a128a8490980aa020f82ca73cc : SUCCESS in 7m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2e9d93dd19ef0d6a0230b01f17e2629d94d1572f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-10-08 23:01:41.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/27/813227/1/check/DellEMC_XtremIO_FC/67d2579/DellEMC_XtremIO_FC/16 : FAILURE in 35m 00s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/27/813227/1/check/DellEMC_XtremIO_iSCSI/f22ab9e/DellEMC_XtremIO_iSCSI/16 : FAILURE in 44m 14s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"3257a037a6a53199ea2c8f4357da8676bb82a26a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-09 01:10:55.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/27/813227/1/check/DellEMC_Unity_FC/cd9331c/DellEMC_Unity_FC/16 : FAILURE in 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/27/813227/1/check/DellEMC_Unity_iSCSI/7ccdf7d/DellEMC_Unity_iSCSI/15 : FAILURE in 23m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"535546270a1208b7780023446a4f71e7ecc3f0ee","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-09 06:59:21.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-813227-1/ns5_nfs : FAILURE in 52m 54s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-813227-1/ns5_iscsi : FAILURE in 1h 35m 27s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"960c1c6becbb273530e449fb44be6abaed4f17e9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-09 08:30:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/27/813227/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d1a8dc2/d1a8dc2.html : TIMED_OUT in 3h 00m 46s","accounts_in_message":[],"_revision_number":1},{"id":"14377869f7010746bcaa7305a87e2df1768a7bda","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-09 08:36:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/27/813227/1/check/iscsi-driver-nimble/dc16c0f/dc16c0f.html : TIMED_OUT in 3h 00m 41s","accounts_in_message":[],"_revision_number":1},{"id":"21ab70a30096e2fe8343a4d06663bea14eb2f22f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-10-10 16:35:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/27/813227/1/check/dsvm-tempest-cisco-zonemanager-job/11ae63d/ : FAILURE in 2h 59m 28s","accounts_in_message":[],"_revision_number":1},{"id":"7b60e89b95b70602c9e14bce5dd64b2b98239acc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-10-12 16:01:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"19eb3a9247f42f3fe532360896894b9400838107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 16:46:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1a39c852dcc84e7d8fd106046dbf2e40 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0553a896773d474fb13dd5cf4243f688 : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc778d5ebe0748baa61e10f0e59288ea : FAILURE in 11m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aaf8fd7e3ba7409fb72c180911c135aa : FAILURE in 10m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5d76b601e9a40829691134f6f2ae67a : FAILURE in 12m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3a5fe9922d74c92b61f19efba6ead29 : SUCCESS in 13m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca69ecbfff5048d780902d30cfad17b0 : SUCCESS in 8m 52s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3d42f6d132234af5ba00d0ee72230d27 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/48860b8d7bd04a29baa28590f0e57494 : SUCCESS in 8m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/17142f24698044f4b5b91cbe5a1513cf : SUCCESS in 9m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c1619cb3ba1a4365ac3f0e29c1fd20b2 : SUCCESS in 35m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f558c95853b64eaba5219b6474d5a946 : FAILURE in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a694c45bb2791d7e70233128b816010d53382357","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 16:53:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a809cf116184a8ead730991cfab7d78 : FAILURE in 19m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c23161e9147f4f9f8f9fbc4f6ab66434 : FAILURE in 24m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6e674439a680e3c92787ee6d0833f3ac35a733c4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-12 18:06:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/27/813227/2/check/iscsi-driver-alletra-6k/f7a1183/f7a1183.html : FAILURE in 53m 19s","accounts_in_message":[],"_revision_number":2},{"id":"cea6c7d7dd96270c0f39215e35acc18277148612","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-10-12 18:43:00.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813227/2/iscsi/ : SUCCESS in 1h 13m 25s\n- cinder-smb http://cloudbase-ci.com/cinder/813227/2/smb/ : SUCCESS in 1h 06m 02s","accounts_in_message":[],"_revision_number":2},{"id":"fa2b92a04227e94650902a2cd6db8f30cf05b273","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-10-12 20:12:41.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/27/813227/2/check/dsvm-tempest-cisco-zonemanager-job/a62875c/ : FAILURE in 30m 12s","accounts_in_message":[],"_revision_number":2},{"id":"578dac8d97756ff96fffe7df359fe861139fa7d6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-12 20:13:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/27/813227/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/7c4b65b/7c4b65b.html : FAILURE in 54m 06s","accounts_in_message":[],"_revision_number":2},{"id":"37ee7a2b6ae92652e423ca2612d9d8b3fb354475","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-12 20:19:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/27/813227/2/check/iscsi-driver-primera/363182d/363182d.html : FAILURE in 50m 05s","accounts_in_message":[],"_revision_number":2},{"id":"099e2933233086fef5c98a124db13423d95be016","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-12 21:08:12.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-813227-2/ns5_nfs : FAILURE in 35m 20s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-813227-2/ns5_iscsi : FAILURE in 32m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"75d4bc26980961b696cfd84a40fa1e57e5ce56b1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-13 02:01:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/27/813227/2/check/iscsi-driver-nimble/85b7ade/85b7ade.html : FAILURE in 57m 50s","accounts_in_message":[],"_revision_number":2},{"id":"d2a4cd3265a51bb105ce6c85807240345d34da1c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-13 07:14:11.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/27/813227/2/check/DellEMC_Unity_FC/5b7a3ca/DellEMC_Unity_FC/35 : FAILURE in 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/27/813227/2/check/DellEMC_Unity_iSCSI/5f2697c/DellEMC_Unity_iSCSI/34 : FAILURE in 23m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"2f1ee80541d67c16b067632ecbcb69d2fbe25135","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-10-22 12:56:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/27/813227/2/20211022165739/NEC-ISCSI : SUCCESS in 2h 38m 00s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/27/813227/2/20211022165739/NEC-FC : SUCCESS in 4h 58m 56s","accounts_in_message":[],"_revision_number":2},{"id":"4a427e00afd11f4b27530ddb080bf8a1d43f68a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-04 16:02:06.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"664d5db53c6b29259def9f32760e71322894ca03","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2021-11-04 16:29:46.000000000","message":"Patch Set 3:\n\nBuild Failed. ibm-gpfs-dsvm-tempest-full: s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/104/index.html","accounts_in_message":[],"_revision_number":3},{"id":"29ee6c8fb35809f2f37f6858e1996856c4fe5e16","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-04 16:44:55.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813227/3/iscsi/ : FAILURE in 14m 04s\n- cinder-smb http://cloudbase-ci.com/cinder/813227/3/smb/ : FAILURE in 13m 58s","accounts_in_message":[],"_revision_number":3},{"id":"3c5179e302bbd6ce909b726fef0e48f8640641fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 16:55:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2418aa29397410bb3d9691a0a9d8b08 : SUCCESS in 17m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4066896ad2084f74bc2e0080e3107784 : SUCCESS in 24m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9d42b5ac07b8e39d49a3323954f5565752f13c19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 17:08:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/aa86a7fce8c742d0a41629c22f21630c : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb2c77dbaed841509d38e798142cc638 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/398900b66e0042d4b9d0cb3265ea0de7 : SUCCESS in 11m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/935276edaeb94038a8ae4ddd88ae6810 : SUCCESS in 12m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39dcf4cc47ac4a7fa3cf86806dec9c0e : SUCCESS in 11m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/544b6085b84141b28b69dea07efc8705 : SUCCESS in 13m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f54a72e6341142ceb4026a464edfb552 : SUCCESS in 9m 17s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/afef6d0ab9434313b9ddbcea418c1e9d : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/08ff8349351c418cb6ef120bd37e797f : SUCCESS in 8m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/22bd6bad04d74e11be5a1d5adfe95d50 : SUCCESS in 8m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bfda00f3a15c492e95b61c52f9639b23 : FAILURE in 35m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a117a71a3d084fe29601077df2062b22 : FAILURE in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c8ea417e31f777ba7b3a3bd1bcbc9087aca0064c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-04 17:52:03.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/27/813227/3/upstream-check/cinder-cDOT-iSCSI/3895d4c/ : FAILURE in 28m 43s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/27/813227/3/upstream-check/cinder-cDOT-NFS/e5a6f20/ : SUCCESS in 41m 12s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/27/813227/3/upstream-check/cinder-cDOT-FCP/930b1d7/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"274ecaa2a172716825f0d41c008b8e4bd269eb31","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-11-04 18:22:01.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/27/813227/3/upstream-solidfire-check/cinder-solidfire-iSCSI/2f6a632/ : SUCCESS in 1h 09m 08s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"6876001deb468eae7385a1ff578bdcc3f863893a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-04 21:16:57.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/27/813227/3/check/iscsi-driver-primera/5ec395a/5ec395a.html : FAILURE in 50m 51s","accounts_in_message":[],"_revision_number":3},{"id":"5099f0dcf29ea413e2e306bf10a7d678219ca311","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-04 22:22:57.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/27/813227/3/check/iscsi-driver-alletra-6k/6e8bdb7/6e8bdb7.html : FAILURE in 52m 15s","accounts_in_message":[],"_revision_number":3},{"id":"42c3be472e6fe18ac999dd9ceacd69e8104fba75","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-11-05 01:16:55.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/27/813227/3/check/DellEMC_PowerFlex/c7f0542/DellEMC_PowerFlex/74 : FAILURE in 1h 52m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"48beda75158a16ade2c9bedf44470baf61d9d92a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-11-05 01:20:31.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/27/813227/3/check/DellEMC_PowerStore_iSCSI/e33bf27/DellEMC_PowerStore_iSCSI/35 : SUCCESS in 1h 47m 14s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/27/813227/3/check/DellEMC_PowerStore_FC/0919dee/DellEMC_PowerStore_FC/42 : SUCCESS in 1h 15m 45s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"794a9a4fcea787d4ac862e0d74876c106c09ad74","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-11-05 02:26:49.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"daa33456b1197829807a3a5a13003be1ede5fd4c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-05 03:09:22.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/27/813227/3/check/iscsi-driver-nimble/0b4fbb8/0b4fbb8.html : FAILURE in 1h 01m 18s","accounts_in_message":[],"_revision_number":3},{"id":"c9a5a892232d9b52f83cd8436e370fd6eb628a50","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-05 04:09:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/27/813227/3/check/dsvm-tempest-cisco-zonemanager-job/6dbfd3f/ : SUCCESS in 1h 08m 38s","accounts_in_message":[],"_revision_number":3},{"id":"64442b8785d18286691ebe8ea72572c3c6532472","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-05 04:34:13.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-813227-3/ns5_nfs : SUCCESS in 4h 48m 58s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-813227-3/ns5_iscsi : SUCCESS in 4h 38m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"379e817fa563f9d1324875f3173b70d8ca4d782f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-06 08:31:24.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/27/813227/3/20211106114108/NEC-ISCSI : SUCCESS in 2h 45m 26s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/27/813227/3/20211106114108/NEC-FC : SUCCESS in 5h 50m 11s","accounts_in_message":[],"_revision_number":3},{"id":"fb6c9a6cfa1a01122c40f3460c88eb7869b02d7b","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-11-07 08:56:18.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/813227-3/Seagate-iSCSI : SUCCESS in 1:22:51\n- Seagate-FC http://os-logs.tristero.net/seagate-ci/813227-3/Seagate-FC : SUCCESS in 49:37.69","accounts_in_message":[],"_revision_number":3},{"id":"2bc8d3cb90261f399f2a96279239971a078344cd","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-11-07 11:17:21.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/813227-3/HPMSA-iSCSI : SUCCESS in 1:11:47\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/813227-3/HPMSA-FC : SUCCESS in 41:26.82","accounts_in_message":[],"_revision_number":3},{"id":"19667c346ff1cd2273ec84fed5384accd1eae4b7","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-11-07 15:24:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/813227-3/Lenovo-iSCSI : SUCCESS in 2:42:31\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/813227-3/Lenovo-FC : SUCCESS in 56:10.04","accounts_in_message":[],"_revision_number":3},{"id":"01ed83fd906f148d81f5c13cb6814f04fa7a0061","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-11-10 00:15:00.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/27/813227/3/check/dsvm-tempest-TOYOU-ACS5000-driver/c0734c4 : FAILURE in 2m 13s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/27/813227/3/check/fc-test/7e4cf9a : FAILURE in 4h 27m 26s","accounts_in_message":[],"_revision_number":3},{"id":"0ad96ba4e5fbb3bc889aec4aec0fea420506493a","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2021-11-15 01:23:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c8329117800603ae391b63f7202ab68e7d880b08","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2021-11-15 01:26:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8457a35b08990d39ac1a987b01954d9ef07f4c71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-15 10:47:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4cc3b9dbc38da4e6e0a8fd08124bd35f2939eb62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 11:16:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c56f7a4ded9d4e6aba17ef769b98409a : FAILURE in 24m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c121d33109b84c50b3732e4060f07549 : SUCCESS in 23m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7fbc3fa1cf6d27f226d4802929a6700af9c9ecf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 11:44:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5a69fe03d46843129e5e3cbee21c9835 : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6c1230722cf4a1d910bab76cc0745cf : SUCCESS in 10m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66f19e13205d45eba9858b0fc9056d9e : SUCCESS in 11m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76e2975e51e7495abbf7fd969b116fb8 : SUCCESS in 12m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/208ca3406f0c4591a7a7024d76fc79f9 : SUCCESS in 15m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fb42ab51b664802bc4d5f9c8e22557e : SUCCESS in 8m 00s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/95a799505e2544d1a3b8d9fd9d7c8dfb : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0b75573501a146578d07f80fd7fdce49 : SUCCESS in 9m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/97c1c0049e3e4b3988f579be2e548c12 : SUCCESS in 11m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a19c9af64cac4817a5df2ef4f61e9d2a : SUCCESS in 50m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3795dab895864cb38fa49735c225af8a : SUCCESS in 8m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"034b3900a018524ab441248a29573f049ec0a699","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-12-15 11:48:58.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"a16a6299343f4bc7d00ed307ac18b932487fae47","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-12-15 12:38:30.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/27/813227/4/upstream-solidfire-check/cinder-solidfire-iSCSI/561cadb/ : FAILURE in 48m 27s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"8929ee96a07601084c18ce839461fe3625fd7b7f","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-12-15 13:12:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-27-813227-4/1639568697 : SUCCESS in 31m 02s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"7373c874aa878b14da106f569d266dafade8589f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-15 13:56:27.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813227/4/iscsi/ : SUCCESS in 1h 04m 42s\n- cinder-smb http://cloudbase-ci.com/cinder/813227/4/smb/ : SUCCESS in 1h 12m 40s","accounts_in_message":[],"_revision_number":4},{"id":"80255de745f7b734ee8c8a17e26d84afdc49f969","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-15 13:56:55.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/27/813227/4/check/iscsi-driver-alletra-6k/eeefcf3/eeefcf3.html : FAILURE in 52m 45s","accounts_in_message":[],"_revision_number":4},{"id":"4c7c0a9890b4b137aa127defa7ea7111d2211201","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-12-15 15:58:26.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/27/813227/4/upstream-check/cinder-cDOT-iSCSI/6017216/ : SUCCESS in 38m 31s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/27/813227/4/upstream-check/cinder-cDOT-NFS/c36f0b1/ : SUCCESS in 39m 20s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/27/813227/4/upstream-check/cinder-cDOT-FCP/15838b0/ : FAILURE in 1h 59m 25s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"fda05b49e6480e419878ec198bf1e053650ca535","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-12-15 17:14:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/27/813227/4/ : SUCCESS in 1h 33m 46s","accounts_in_message":[],"_revision_number":4},{"id":"2830ec73dcd03836665d37b7f088054c8568f1af","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-15 21:43:46.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-813227-4/ns5_nfs : SUCCESS in 4h 54m 50s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-813227-4/ns5_iscsi : SUCCESS in 4h 55m 53s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"3b9f2519b10c478dd9b7644dd7ecbbbfcf164ed2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-15 22:21:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a47b7d968c7240ce89242256c7788b20 : SUCCESS in 1h 26m 08s","accounts_in_message":[],"_revision_number":4},{"id":"b5dbcf69aafbbf7a02773d4065a87ff6a8ea3851","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-15 23:10:02.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/27/813227/4/check/iscsi-driver-nimble/01016db/01016db.html : FAILURE in 1h 00m 05s","accounts_in_message":[],"_revision_number":4},{"id":"51ec1d30aeaec0b61bc5d29c7c74fbeb8ff9ec0e","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-16 01:11:18.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/27/813227/4/check/dsvm-tempest-TOYOU-ACS5000-driver/71dadf8 : FAILURE in 1h 24m 05s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/27/813227/4/check/fc-test/4c58cda : FAILURE in 1h 14m 38s","accounts_in_message":[],"_revision_number":4},{"id":"acbf0e3f34d3989baae367ff424b6bd4ee3eaea6","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-12-17 12:04:19.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/27/813227/4/20211217180716/NEC-ISCSI : SUCCESS in 2h 39m 47s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/27/813227/4/20211217180716/NEC-FC : SUCCESS in 2h 56m 55s","accounts_in_message":[],"_revision_number":4},{"id":"6916a0062fb0eeaa183fb3d6190fc8b330f24445","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-01-31 18:17:06.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"445c910b8a8cee713a07493c9236c6c2dc04588e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-01-31 18:23:16.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"5dac52b4e8afa1272d38d14573edae7ffca54514","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 19:30:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/54fcc432388d4a9bbdadd4425a8c3a97 : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/056e8cf818444cc4995f3aba8089f997 : SUCCESS in 9m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/380fbc6054324f67bce307f4c463b0ce : SUCCESS in 11m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b5144e2a9c6435d9b955879ac92a68b : SUCCESS in 12m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b071cf67644541898b58de5c7c4db621 : SUCCESS in 13m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bbd455577684f998ca129ad045fc448 : SUCCESS in 7m 49s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a63a516738cc43258ccfbfa0cf3ec23c : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5596f8c848ac48a3b6129fb28b1eadcf : SUCCESS in 9m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ae67dceae6b940be92707389a1a0a857 : SUCCESS in 7m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1d0bdb74d6d34dce9259f0783426c085 : SUCCESS in 34m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/73e4069a123647e29945a37d385298be : SUCCESS in 6m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4e05c5ae5cba169369289d1fa1dae387d7106241","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-31 19:38:53.000000000","message":"Patch Set 5:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/31_01_2022_19:30:55\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"e06dd560815c719ed75bb7516737470ca949dd21","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-31 19:48:01.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/27/813227/5/check/DellEMC_PowerFlex/ee5007b/DellEMC_PowerFlex/339 : FAILURE in 16m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"6ade52b234cd1bf0b9674817c253d581a7176aef","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-01-31 19:59:19.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/27/813227/5/upstream-solidfire-check/cinder-solidfire-iSCSI/0f5dfe3/ : FAILURE in 24m 58s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"15a3ddf2617ae2ab07e1e1548518b3cee0681212","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 20:25:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c04f391ebfd49bf8618374862526a0a : FAILURE in 35m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/074d09fa01ff4ae1a7961ee44af0f1bd : FAILURE in 36m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f02670038a000f612674d6ddf93e8d928ace3006","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-31 20:40:30.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813227/5/iscsi/ : SUCCESS in 1h 04m 28s\n- cinder-smb http://cloudbase-ci.com/cinder/813227/5/smb/ : SUCCESS in 50m 24s","accounts_in_message":[],"_revision_number":5},{"id":"3eef1fbea9adb7fdd4ff6cf8ca283267b1c3daa9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-31 20:50:17.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/27/813227/5/check/DellEMC_PowerFlex/1f094a9/DellEMC_PowerFlex/344 : FAILURE in 23m 57s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"4e98ea9e9106ed045838f579246cc94cd0f40353","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-01-31 21:20:50.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-27-813227-5/1643657449 : SUCCESS in 32m 31s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":5},{"id":"66dd7a3ada6efaa2fd9f8a280011857c7a91700b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-01-31 21:38:10.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/27/813227/5/upstream-solidfire-check/cinder-solidfire-iSCSI/3b85a57/ : SUCCESS in 1h 06m 10s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"4bec95c65f76048b1efe4fc3d8ba0f1bb42918ae","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-01-31 22:07:23.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/27/813227/5/check/iscsi-driver-alletra-6k/77fba56/77fba56.html : FAILURE in 21m 48s","accounts_in_message":[],"_revision_number":5},{"id":"bd49728b0f0581e063a6fe5eed85786a880b19f7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-01-31 22:46:11.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/27/813227/5/check/iscsi-driver-primera/41ec8b4/41ec8b4.html : FAILURE in 50m 10s","accounts_in_message":[],"_revision_number":5},{"id":"d97f719949770e5628fb26fc74eeb0f5c1ef92d6","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-01-31 23:20:57.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/27/813227/5/check/DellEMC_PowerMAX_FC/6e21dcc/DellEMC_PowerMAX_FC/253 : FAILURE in 52m 50s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/27/813227/5/check/DellEMC_PowerMAX_iSCSI/cb30d23/DellEMC_PowerMAX_iSCSI/266 : FAILURE in 54m 38s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"3fb79d1703bd287ff3a9e4e5b193c6d45e46a2b0","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-01-31 23:23:29.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/27/813227/5/check/DellEMC_PowerStore_iSCSI/c916158/DellEMC_PowerStore_iSCSI/292 : SUCCESS in 1h 19m 29s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/27/813227/5/check/DellEMC_PowerStore_FC/d389b8e/DellEMC_PowerStore_FC/309 : SUCCESS in 1h 05m 14s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"c75aa91c0afc5557dde8a474d9e2646fcfb86edc","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-31 23:39:33.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/27/813227/5/upstream-check/cinder-cDOT-iSCSI/0df7bcf/ : FAILURE in 7m 58s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/27/813227/5/upstream-check/cinder-cDOT-NFS/03a6062/ : FAILURE in 4h 06m 32s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/27/813227/5/upstream-check/cinder-cDOT-FCP/1e5bc95/ : SUCCESS in 1h 55m 55s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"c7fa569e0e4a88c03024ca08ee67dc800d219975","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-01 00:32:02.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-27-813227-5/1643660748 : SUCCESS in 29m 18s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":5},{"id":"1572a21ffb00585bee665d335939997c1bd7c753","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-01 01:48:13.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/27/813227/5/check/dsvm-tempest-cisco-zonemanager-job/6a58e0d/ : POST_FAILURE in 14m 32s","accounts_in_message":[],"_revision_number":5},{"id":"870a3748f91af285f3a875fe46662a2c14eb012d","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-02-01 03:12:00.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/27/813227/5/check/DellEMC_SC_FC/d212be2/DellEMC_SC_FC/276 : FAILURE in 1h 08m 49s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/27/813227/5/check/DellEMC_SC_iSCSI/1d6f959/DellEMC_SC_iSCSI/272 : SUCCESS in 1h 44m 51s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"9d6339529e4479f66c762f42ec6e8d15a0d9cbee","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-01 04:35:45.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/813227-5/HPMSA-iSCSI/ : SUCCESS in 52:17.78\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/813227-5/HPMSA-FC/ : SUCCESS in 43:04.15","accounts_in_message":[],"_revision_number":5},{"id":"6757f21d7364ee15d4ce88caab3f5b05d546411e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-01 05:10:21.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-813227-5/ns5_nfs : SUCCESS in 4h 29m 30s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-813227-5/ns5_iscsi : SUCCESS in 4h 31m 23s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"bf2738b6fef1d54f5505d4a5d63cba67d85da740","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-01 06:15:02.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/27/813227/5/check/iscsi-driver-nimble/2475287/2475287.html : FAILURE in 58m 09s","accounts_in_message":[],"_revision_number":5},{"id":"57bbfe7fb59a0ab67d0c5763ba4972a96fbf0bfc","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-01 06:39:05.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/813227-5/Lenovo-iSCSI/ : SUCCESS in 58:58.82\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/813227-5/Lenovo-FC/ : SUCCESS in 56:20.23","accounts_in_message":[],"_revision_number":5},{"id":"cf9828f71f20737babd91aa8e908f44ff7a17390","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-02-01 07:13:18.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/27/813227/5/check/DellEMC_Unity_FC/4942f73/DellEMC_Unity_FC/496 : FAILURE in 1h 21m 44s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/27/813227/5/check/DellEMC_Unity_iSCSI/b50de92/DellEMC_Unity_iSCSI/490 : FAILURE in 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"eb3a95d09b445d53f5029abb7bf86ea8938b878f","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-01 07:16:35.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/813227-5/PVME-iSCSI/ : SUCCESS in 1:01:49\n- PVME-FC https://os-logs.seagate.com/dell-ci/813227-5/PVME-FC/ : SUCCESS in 1:01:38","accounts_in_message":[],"_revision_number":5},{"id":"5e4254899cc36924891e8904597ad03d8eeef7ba","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-02-01 07:43:34.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/813227-5/Seagate-iSCSI/ : FAILURE in 1:13:00\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/813227-5/Seagate-FC/ : SUCCESS in 1:02:13","accounts_in_message":[],"_revision_number":5},{"id":"c10972af29a0ba3ecbfbddf40087d6b821f61dff","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-01 09:59:23.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/27/813227/5/check/DellEMC_XtremIO_FC/b60444c/DellEMC_XtremIO_FC/610 : SUCCESS in 59m 03s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/27/813227/5/check/DellEMC_XtremIO_iSCSI/561d417/DellEMC_XtremIO_iSCSI/620 : SUCCESS in 1h 17m 44s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"ba1914f41c6a087126d5a212ff9745c3ebec9155","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-02-01 20:42:53.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/27/813227/5/20220202031532/NEC-V-FC : SUCCESS in 2h 05m 03s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/27/813227/5/20220202031532/NEC-V-ISCSI : SUCCESS in 2h 27m 18s","accounts_in_message":[],"_revision_number":5},{"id":"da236a2eddc07728b3d7e0c776372f42c29fbb3f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-03 03:18:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/27/813227/5/ : SUCCESS in 1h 48m 42s","accounts_in_message":[],"_revision_number":5},{"id":"2b5e47d495ac73e49c6a787b177596688f6471e3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-02-04 21:16:19.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/27/813227/5/20220205025336/NEC-FC : SUCCESS in 3h 17m 26s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/27/813227/5/20220205025336/NEC-ISCSI : SUCCESS in 3h 22m 36s","accounts_in_message":[],"_revision_number":5},{"id":"f970bf59630c9cc4b3cb56e1e5293dcaf7b3b7b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-20 19:58:21.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4c14eb595a712ec1a529724b8deef3ed58982aaf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 20:38:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/284d99bca3be4563ad73cc761e43b4ac : FAILURE in 30m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1941764255ec43c1bbd4bf6a25974e59 : FAILURE in 31m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e1f5f2093d46d50fbfc016deab64ef11fcdcb0b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 20:46:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/aee3cf1d8bf9463da5c63c8277482220 : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/106d0609dadc442e9f71bc4068c4594c : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7780bb3b0035473f94687a6be020a311 : SUCCESS in 8m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0435660899eb4e819a1e9185e4f4f30f : SUCCESS in 15m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2b32f04ddfd40f58587fbdee2218233 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9a43f2e60444ec986cc6f28e5198a90 : SUCCESS in 6m 57s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bfb15f7c139c4114a9ec5de724527f5c : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a553ca44f5b44f1295918430a736a75d : SUCCESS in 6m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/239aee7f4ba145b8b5c8a4ba05b7824c : SUCCESS in 7m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b9f4c5dfc8b44125beec91b74a15b5ed : SUCCESS in 41m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/85fdce5e98414c4ea5fdb05e499eb6ac : SUCCESS in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dff2a26bfcbf04a880a8d0f6342fa6af8cc45643","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-20 21:00:30.000000000","message":"Patch Set 6:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15222/","accounts_in_message":[],"_revision_number":6},{"id":"238cfacaeb374c59f0a0bcf31ba6f902c6efdb95","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-20 21:02:12.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15222/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"c8ceab58937630dbfdeab2cc5f4c21c45d3f67b3","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-20 21:26:06.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-27-813227-6/1645389994 : SUCCESS in 17m 22s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":6},{"id":"5f6e96d7125df3c7c40492af6dc13be3255fcbcc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-21 01:49:26.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/813227/6/iscsi/ : FAILURE in 1h 33m 01s\n- cinder-smb http://cloudbase-ci.com/cinder/813227/6/smb/ : SUCCESS in 1h 03m 14s","accounts_in_message":[],"_revision_number":6},{"id":"d7d5a81b801928386a6451a593b07723f5fdfa88","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-21 03:44:03.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/813227-6/HPMSA-iSCSI/ : SUCCESS in 50:34.92\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/813227-6/HPMSA-FC/ : SUCCESS in 44:04.93","accounts_in_message":[],"_revision_number":6},{"id":"ad8b65a2e7c9a6844d93d6aae901f5520b7dd843","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-21 05:23:53.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/813227-6/Lenovo-iSCSI/ : SUCCESS in 1:00:50\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/813227-6/Lenovo-FC/ : SUCCESS in 57:23.97","accounts_in_message":[],"_revision_number":6},{"id":"5579de5d4c28ef58f9d7d370d0733cce3aa348d5","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-21 05:41:59.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/813227-6/PVME-iSCSI/ : SUCCESS in 1:04:56\n- PVME-FC https://os-logs.seagate.com/dell-ci/813227-6/PVME-FC/ : SUCCESS in 1:01:16","accounts_in_message":[],"_revision_number":6},{"id":"89bd6122c86af2995338e079df9ffbfb025996e7","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-02-21 06:04:27.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/813227-6/Seagate-iSCSI/ : SUCCESS in 1:03:43\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/813227-6/Seagate-FC/ : SUCCESS in 1:04:04","accounts_in_message":[],"_revision_number":6},{"id":"e60599cef494840620d470a599795d5cc2ac61ed","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-21 09:45:40.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/27/813227/6/check/dsvm-tempest-cisco-zonemanager-job/b7b22b4/ : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":6},{"id":"477b607998782971f0ee297b3f63a01df5681158","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-21 11:29:02.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/27/813227/6/check/iscsi-driver-alletra-6k/03a52e2/03a52e2.html : FAILURE in 50m 20s","accounts_in_message":[],"_revision_number":6},{"id":"496b9c6807935ad23f242f858cf53087211bcb75","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-21 15:46:00.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-813227-6/ns5_nfs : FAILURE in 31m 11s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-813227-6/ns5_iscsi : SUCCESS in 4h 55m 12s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"2083a101467065bcf1363579fa466efa9cc59cf9","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2022-02-21 21:00:06.000000000","message":"Patch Set 6:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/27/813227/6/fc : SUCCESS in 1h 09m 24s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/27/813227/6/iscsi : FAILURE in 0h 25m 38s","accounts_in_message":[],"_revision_number":6},{"id":"b9bb2c6311b404dbce89da5b13624cfcde0109a0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-02-22 01:22:54.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/27/813227/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/aac4eeb/aac4eeb.html : FAILURE in 1h 18m 12s","accounts_in_message":[],"_revision_number":6},{"id":"2f0d8a21315871a1e7e95add17341ad6ec072add","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-23 02:01:53.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/27/813227/6/ : SUCCESS in 1h 34m 58s","accounts_in_message":[],"_revision_number":6},{"id":"d12d40206d15e97e0242cd961834d915e241aae9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-02-24 18:44:41.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/27/813227/6/20220224210409/NEC-FC : SUCCESS in 3h 29m 42s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/27/813227/6/20220224210409/NEC-ISCSI : SUCCESS in 6h 40m 29s","accounts_in_message":[],"_revision_number":6},{"id":"594fe2fcc9f0a9c7045002e3d7c9b2dacfa3cbd4","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-02-25 23:08:34.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/27/813227/6/20220226054040/NEC-V-FC : SUCCESS in 2h 13m 03s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/27/813227/6/20220226054040/NEC-V-ISCSI : SUCCESS in 2h 27m 51s","accounts_in_message":[],"_revision_number":6},{"id":"0405f28d2f9d3c0895653a4ce8d2d8145ddd77a0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-01 04:01:07.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1918f105b980b0ee6802bc477ac6be2db683bc6b","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-03-02 20:12:47.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"134f40af5cd03f61975e25c846ff49abc5479169","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-03 13:58:01.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d982df2130c566c7e73b96d14b4f3ee80d3a9bd0","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-03-03 14:31:28.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4512ba131422b82d42244d51531457f7779351b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 14:22:33.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"03e9fe195f56769f5032f1270722502a33e08a52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 15:45:44.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0354792b30e44ad5884b35ce1337c820 : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4e9b125cb654473bf362b65c9f4bd0c : SUCCESS in 9m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d5be90bd41246f38c7b549df467c212 : SUCCESS in 13m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/019b01f6f018463a9343390f26f6df47 : SUCCESS in 15m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ee656749cdf4f65894f6a6d16d4666e : SUCCESS in 14m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf552ac746db4cd0ba6e136e0ecd0eac : SUCCESS in 9m 42s","accounts_in_message":[],"_revision_number":6},{"id":"aa0664f83be6a07556c2e4e336f0f8593666be61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 15:45:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"f107313a4f76bd8949ed8d295ea7724194399cf4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 15:53:18.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/c86abfdc8b12499bbe5796f561b0a4d3 : SUCCESS in 1m 26s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb10b7c7f7cf4bbd9abf29e1b08b83f1 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb971f048b8f4749b8247a7e477c9341 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"52370e58fe107d2bc2de6df53305707229be21d7","revisions":{"0b8dfbe27999be7ac24898e3959ab80850fada0e":{"kind":"REWORK","_number":1,"created":"2021-10-08 16:38:33.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/27/813227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/813227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/813227/1"}}},"commit":{"parents":[{"commit":"c035a0b13a8646b0e91ff63937276101d9e4cb3d","subject":"db: Enable auto-generation of database migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c035a0b13a8646b0e91ff63937276101d9e4cb3d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-24 17:25:58.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-10-08 16:38:26.000000000","tz":60},"subject":"docs: Remove unnecessary \u0027blockstorage-\u0027 prefix","message":"docs: Remove unnecessary \u0027blockstorage-\u0027 prefix\n\nThese are in the cinder doc tree so of course they\u0027re block\nstorage-related.\n\nChange-Id: Ic1950ff89021a89de397619eef17f8100eb3d847\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b8dfbe27999be7ac24898e3959ab80850fada0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b8dfbe27999be7ac24898e3959ab80850fada0e"}]},"branch":"refs/heads/master"},"ebe402e51a17cea816b0610228773bda3529e55e":{"kind":"REWORK","_number":2,"created":"2021-10-12 16:01:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/27/813227/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/813227/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/813227/2"}}},"commit":{"parents":[{"commit":"e80929939d5c4c47555faf7fe074f0b2604579d4","subject":"db: Enable auto-generation of database migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e80929939d5c4c47555faf7fe074f0b2604579d4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-24 17:25:58.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-10-12 16:01:16.000000000","tz":60},"subject":"docs: Remove unnecessary \u0027blockstorage-\u0027 prefix","message":"docs: Remove unnecessary \u0027blockstorage-\u0027 prefix\n\nThese are in the cinder doc tree so of course they\u0027re block\nstorage-related.\n\nChange-Id: Ic1950ff89021a89de397619eef17f8100eb3d847\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ebe402e51a17cea816b0610228773bda3529e55e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ebe402e51a17cea816b0610228773bda3529e55e"}]},"branch":"refs/heads/master"},"f18eb25c4810c3d3035d3c6c925bcfecc7d0e54c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-11-04 16:02:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/27/813227/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/813227/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/813227/3"}}},"commit":{"parents":[{"commit":"4b2c251765ee1055b37499e6f5c6c7bd584290f7","subject":"db: Enable auto-generation of database migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4b2c251765ee1055b37499e6f5c6c7bd584290f7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-24 17:25:58.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-11-04 12:44:29.000000000","tz":0},"subject":"docs: Remove unnecessary \u0027blockstorage-\u0027 prefix","message":"docs: Remove unnecessary \u0027blockstorage-\u0027 prefix\n\nThese are in the cinder doc tree so of course they\u0027re block\nstorage-related.\n\nChange-Id: Ic1950ff89021a89de397619eef17f8100eb3d847\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f18eb25c4810c3d3035d3c6c925bcfecc7d0e54c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f18eb25c4810c3d3035d3c6c925bcfecc7d0e54c"}]},"branch":"refs/heads/master"},"a7b88ca9fd5e53ac251cb015ec4566a2f1b29335":{"kind":"REWORK","_number":4,"created":"2021-12-15 10:47:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/27/813227/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/813227/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/813227/4"}}},"commit":{"parents":[{"commit":"b770e5c15016d81290539180625482e390348dfe","subject":"db: Enable auto-generation of database migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b770e5c15016d81290539180625482e390348dfe"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-24 17:25:58.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-12-15 10:46:52.000000000","tz":0},"subject":"docs: Remove unnecessary \u0027blockstorage-\u0027 prefix","message":"docs: Remove unnecessary \u0027blockstorage-\u0027 prefix\n\nThese are in the cinder doc tree so of course they\u0027re block\nstorage-related.\n\nChange-Id: Ic1950ff89021a89de397619eef17f8100eb3d847\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7b88ca9fd5e53ac251cb015ec4566a2f1b29335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7b88ca9fd5e53ac251cb015ec4566a2f1b29335"}]},"branch":"refs/heads/master"},"5e4945fe8558532a60ba2c86a379f4f54d8b37e9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-01-31 18:17:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/27/813227/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/813227/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/813227/5"}}},"commit":{"parents":[{"commit":"c000282318e037e11c57178c88de76a28d4b36cb","subject":"db: Enable auto-generation of database migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c000282318e037e11c57178c88de76a28d4b36cb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-24 17:25:58.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-01-31 18:16:55.000000000","tz":0},"subject":"docs: Remove unnecessary \u0027blockstorage-\u0027 prefix","message":"docs: Remove unnecessary \u0027blockstorage-\u0027 prefix\n\nThese are in the cinder doc tree so of course they\u0027re block\nstorage-related.\n\nChange-Id: Ic1950ff89021a89de397619eef17f8100eb3d847\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5e4945fe8558532a60ba2c86a379f4f54d8b37e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5e4945fe8558532a60ba2c86a379f4f54d8b37e9"}]},"branch":"refs/heads/master"},"52370e58fe107d2bc2de6df53305707229be21d7":{"kind":"REWORK","_number":6,"created":"2022-02-20 19:58:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/27/813227/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/813227/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/813227/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/813227/6"}}},"commit":{"parents":[{"commit":"31f8ad4eb4279d356c67e9f14531ce2ca6c70ea5","subject":"db: Enable auto-generation of database migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31f8ad4eb4279d356c67e9f14531ce2ca6c70ea5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-24 17:25:58.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-02-20 19:06:32.000000000","tz":0},"subject":"docs: Remove unnecessary \u0027blockstorage-\u0027 prefix","message":"docs: Remove unnecessary \u0027blockstorage-\u0027 prefix\n\nThese are in the cinder doc tree so of course they\u0027re block\nstorage-related.\n\nChange-Id: Ic1950ff89021a89de397619eef17f8100eb3d847\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/52370e58fe107d2bc2de6df53305707229be21d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/52370e58fe107d2bc2de6df53305707229be21d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
