)]}'
{"id":"openstack%2Fcinder~815582","triplet_id":"openstack%2Fcinder~master~I6e4ebcb9dc67e444157aaa85fcd0239817a79136","project":"openstack/cinder","branch":"master","topic":"bp/hitachi-vsp-add-multi-pool","attention_set":{},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2023-01-27 19:02:55.000000000","reason":"Change was submitted"},"33473":{"account":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"last_update":"2023-01-27 19:02:55.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-01-27 19:02:55.000000000","reason":"Change was submitted"},"34162":{"account":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"last_update":"2022-12-16 02:38:16.000000000","reason":"removed on reply"},"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2022-11-02 00:19:36.000000000","reason":"\u003cGERRIT_ACCOUNT_34148\u003e replied on the change","reason_account":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"}},"34148":{"account":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"last_update":"2023-01-20 05:46:30.000000000","reason":"\u003cGERRIT_ACCOUNT_34148\u003e replied on the change","reason_account":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"}},"13671":{"account":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"last_update":"2022-10-31 08:35:51.000000000","reason":"\u003cGERRIT_ACCOUNT_13671\u003e replied on the change","reason_account":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"}},"31721":{"account":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"last_update":"2023-01-27 19:02:55.000000000","reason":"Change was submitted"},"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2022-07-28 05:56:25.000000000","reason":"\u003cGERRIT_ACCOUNT_30555\u003e replied on the change","reason_account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2023-01-27 19:02:55.000000000","reason":"Change was submitted"},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2023-01-25 10:00:40.000000000","reason":"\u003cGERRIT_ACCOUNT_20813\u003e replied on the change","reason_account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-01-24 17:25:37.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I6e4ebcb9dc67e444157aaa85fcd0239817a79136","subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","status":"MERGED","created":"2021-10-27 05:00:12.000000000","updated":"2023-01-31 20:17:31.000000000","submitted":"2023-01-27 19:02:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":100,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"815582-bp/hitachi-vsp-add-multi-pool","meta_rev_id":"3245713ca53d4bdeba782e73110185e27224e0c6","_number":815582,"virtual_id_number":815582,"owner":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-27 19:02:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"all":[{"value":1,"date":"2022-12-26 23:33:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":2,"date":"2023-01-24 22:16:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":1,"date":"2022-10-31 08:35:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":-1,"date":"2023-01-20 16:51:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":2,"date":"2023-01-25 10:00:40.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2023-01-24 17:25:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":1,"date":"2023-01-25 10:00:40.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":1,"date":"2023-01-24 17:25:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-27 05:00:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-27 05:32:08.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-10-27 05:32:49.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-27 06:09:50.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-10-27 06:29:17.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-27 07:03:32.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-10-27 07:14:26.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-27 07:50:42.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-10-27 09:50:38.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-27 14:49:21.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-28 07:15:45.000000000","updated_by":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"reviewer":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"state":"CC"},{"updated":"2021-11-25 05:15:57.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2021-11-25 08:24:45.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-11-25 08:24:50.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 08:51:00.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 08:58:49.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-11-25 09:31:55.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 10:21:57.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-11-25 10:25:16.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-11-25 17:12:29.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-26 09:56:42.000000000","updated_by":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"reviewer":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"state":"CC"},{"updated":"2021-11-26 17:32:41.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-12-06 07:44:16.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-12 15:24:53.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2022-01-17 08:59:50.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-01-17 11:42:30.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-17 12:20:53.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-17 12:52:33.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-01-17 15:14:32.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-01-20 14:07:22.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-01-21 09:11:09.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-01-21 09:52:12.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2022-01-21 13:03:38.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-01-21 13:07:39.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-01-21 13:34:15.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-06-23 00:48:57.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-06-23 16:33:56.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-07-18 22:56:05.000000000","updated_by":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"reviewer":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"state":"CC"},{"updated":"2022-07-27 06:31:19.000000000","updated_by":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"reviewer":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"state":"REVIEWER"},{"updated":"2022-07-28 00:48:29.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-07-28 02:00:27.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-17 05:46:22.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2022-10-18 16:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2022-10-20 14:06:53.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"state":"CC"},{"updated":"2022-10-26 19:19:56.000000000","updated_by":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"reviewer":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"state":"CC"},{"updated":"2022-10-31 08:35:51.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"state":"REVIEWER"},{"updated":"2022-11-30 14:26:01.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2023-01-03 15:20:43.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2023-01-11 14:23:33.000000000","updated_by":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"reviewer":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"state":"REVIEWER"},{"updated":"2023-01-20 15:02:23.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2023-01-24 17:25:37.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"97431c10e72c435fc8d4bdd46a29207c04596c78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2021-10-27 05:00:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80689a304b7f2196cb6ce861d1f9edcdc7f2e725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 05:00:57.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"6d710e3ffb61a2ec602bc64b388d362f20d3011d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 05:01:03.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"7a6d3a4060fe08fc99eed097d2eadc7b5d25bbea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2021-10-27 05:06:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2d2dcc913c1eeb0a54ca6dc13a42574eae59d689","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 05:30:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/27bb041d326146c7aaa84c62d62e8a3c : SUCCESS in 16m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a26d1804bda473b9a0555cec9bbcecd : SUCCESS in 21m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cd6c74b13fb8af6500792157702dd2d6083c8874","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-27 05:32:08.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/2/check/DellEMC_Unity_FC/94f6232/DellEMC_Unity_FC/87 : FAILURE in 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/2/check/DellEMC_Unity_iSCSI/d6c0cbe/DellEMC_Unity_iSCSI/83 : FAILURE in 23m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"2ab057e7d6f04cfd0098031c31a7ce8d0d226e08","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-10-27 05:32:49.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"b0bbd41a47768e4f95d94c49e7e553b7d7d16769","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-10-27 06:09:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/2/iscsi/ : SUCCESS in 1h 03m 01s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/2/smb/ : SUCCESS in 46m 25s","accounts_in_message":[],"_revision_number":2},{"id":"e3c467c75c1b56456fcf9413efe064d04bf80b6d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-27 06:29:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/82/815582/2/check/iscsi-driver-primera/484d135/484d135.html : SUCCESS in 1h 18m 19s","accounts_in_message":[],"_revision_number":2},{"id":"3f44ff3155dae5d72f48f7a625600694bea04406","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-27 06:50:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/815582/2/check/iscsi-driver-alletra-6k/6ee7df0/6ee7df0.html : SUCCESS in 1h 39m 37s","accounts_in_message":[],"_revision_number":2},{"id":"d3ad951cfcd7740607848590c02d8c08817ed050","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-10-27 07:03:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/82/815582/2/check/dsvm-tempest-cisco-zonemanager-job/6ffff49/ : FAILURE in 1h 00m 47s","accounts_in_message":[],"_revision_number":2},{"id":"fc4194153efe19e3645eb2e4262a839c1f4b37c0","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-10-27 07:14:26.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/82/815582/2/check/DellEMC_XtremIO_FC/89223f1/DellEMC_XtremIO_FC/48 : SUCCESS in 1h 02m 09s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/2/check/DellEMC_XtremIO_iSCSI/f543f5d/DellEMC_XtremIO_iSCSI/46 : FAILURE in 28m 50s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"0a0d628333a6f620314429849187912e4f601282","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-27 07:42:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/815582/2/check/iscsi-driver-nimble/d2b41ab/d2b41ab.html : SUCCESS in 1h 51m 14s","accounts_in_message":[],"_revision_number":2},{"id":"f924ab927bee6a9c1cd7de5a3841f801136b41fc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-10-27 07:50:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5a00484/ : SUCCESS in 2h 25m 40s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/27cd716199624649b91513bbfae2dd3f : DISK_FULL in 2h 43m 20s","accounts_in_message":[],"_revision_number":2},{"id":"a79992e91bbc7c6155cea985451aefa496a28219","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 07:50:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2247ebf961f448a89afe4bc5161000df : FAILURE in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d188268798d47a3a7342f31b8ea3b8e : SUCCESS in 9m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3380b9da03147fe974d2d69544ec60f : SUCCESS in 10m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/953c30b7cb564227a3f8131ce287de8b : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fcba095d20f4594af75a45f0c493fb0 : FAILURE in 6m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/39e1bfb590b749d2899b0f9747056c2a : SUCCESS in 55m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a95f20063d7d48f684f6c7cea9d28f54 : SUCCESS in 1h 03m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4e13c6841fea42b09dcee4b1ded4e83a : SUCCESS in 14m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b033df2248b844d7a4dca7e74a8f3c60 : SUCCESS in 6m 12s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/58fe7e9d1fce44f08f0580dff00b0bcc : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7234e65ea1b649b1b88f806fe16c7be4 : SUCCESS in 6m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/31982d2ed4494b80aadb3c017c23f733 : SUCCESS in 7m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/95eaad566dc148b0948a82f515811a6c : SUCCESS in 38m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8f39de7c653b4017b777ece1becba136 : FAILURE in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2021d48a1d634d08ba1eb2bf0270591f : SUCCESS in 1h 24m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7e8453efd14645a79444e9bae00ef836 : SUCCESS in 1h 38m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/46ee673338364d4ca5fde64e2b361bac : SUCCESS in 1h 06m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7428fe67911741dabd09e16f7339a55f : SUCCESS in 1h 12m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1b031dcc194540e090bcc405f76717c5 : SUCCESS in 1h 15m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5f0ae51ad7ce4b4dad6fac406e0ba36d : SUCCESS in 39m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1a53e09a35f74f81a662879b2f16df86 : SUCCESS in 1h 08m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3e000b2c26d34004b86d18894721263e : SUCCESS in 2h 36m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49a22766ae634d7c8d50bb232743e5b1 : SUCCESS in 50m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38a8657d19de4e6e878291dfba17e82b : SUCCESS in 50m 36s","accounts_in_message":[],"_revision_number":2},{"id":"56a32e6734893ea1be5b5d33034e810959754489","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-27 09:50:38.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-815582-1/ns5_nfs : SUCCESS in 4h 16m 18s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-815582-1/ns5_iscsi : SUCCESS in 4h 20m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"f4d7b72a5f33fd3995709af55018e14b216b7881","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-27 10:31:34.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-815582-2/ns5_nfs : SUCCESS in 4h 12m 5s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-815582-2/ns5_iscsi : SUCCESS in 4h 16m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"0134203e8e934b4f528d2607c5930364fe88c05a","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-10-27 14:49:21.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/2/20211027204404/NEC-FC : SUCCESS in 2h 50m 38s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/2/20211027204404/NEC-ISCSI : SUCCESS in 3h 05m 14s","accounts_in_message":[],"_revision_number":2},{"id":"e4bbd16b3b75e48cff2095d7df1c6f58e796c14a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2021-10-28 07:15:45.000000000","message":"Patch Set 1:\n\nBuild Successful. ibm-gpfs-dsvm-tempest-full: https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/56/index.html","accounts_in_message":[],"_revision_number":1},{"id":"b0d45add6f840c8dde8e4ecec82c4e98f80ea107","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2021-10-28 08:39:07.000000000","message":"Patch Set 2:\n\nBuild Failed. ibm-gpfs-dsvm-tempest-full: s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/57/index.html","accounts_in_message":[],"_revision_number":2},{"id":"eabe7cb4b99f550d837e77f1a61f18dcc0a9c8ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2021-10-28 11:13:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"36d37cc454371b3a709110f0b9a75da68b9b51fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 11:42:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74165c19a4174e09891ff5b8b98cc0cf : SUCCESS in 22m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe560bf7a0d348e6b20fbfac3ef11782 : SUCCESS in 23m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"be8e5de470858567f9ec97deb401af4730dd5331","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-10-28 12:10:33.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/3/iscsi/ : FAILURE in 39m 29s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/3/smb/ : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":3},{"id":"5b128c27ceb17a01258bb0f257fb914385a2f588","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-28 12:19:21.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/815582/3/check/iscsi-driver-alletra-6k/9349d12/9349d12.html : FAILURE in 1h 04m 04s","accounts_in_message":[],"_revision_number":3},{"id":"0d8bc3984d66f686d026ad832f5ac90a2f768e81","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-28 12:43:59.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/82/815582/3/check/iscsi-driver-primera/c56c489/c56c489.html : FAILURE in 1h 27m 37s","accounts_in_message":[],"_revision_number":3},{"id":"4b6c4e8b8981566f04707df7126716d96dec5f3f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-10-28 13:53:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/82/815582/3/check/dsvm-tempest-cisco-zonemanager-job/834f733/ : SUCCESS in 1h 07m 30s","accounts_in_message":[],"_revision_number":3},{"id":"ef0e9ee05f000dfe6bf6c6396c55c9d9c092c211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 14:09:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b30837735f5e4c1692b5b5e863d30fb8 : FAILURE in 9m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83fd2c12130f45f49d14afc45a137672 : SUCCESS in 11m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f40295c9c98476c882f05f29af8c683 : SUCCESS in 13m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/634d96bc0ea04db88a1f9e5fbe4662ea : SUCCESS in 14m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/523963e9863d4c148dd45a04136e8928 : SUCCESS in 14m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/faea63779c974e0c8ddb6457ba628c40 : SUCCESS in 57m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cc38d0eba6e64b31933b66547791efd6 : SUCCESS in 1h 12m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/533c0db8d7724f309e8569d7bf9a41ad : SUCCESS in 14m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e7513751562943e394fc4b2d4afefed2 : SUCCESS in 7m 42s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ed1fe1d6fc674da1ace3c8f8f1316fbb : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/34e7c702bc604bbaacf95c6851547cd4 : SUCCESS in 7m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5476dee5a35045a5a86b7657fa2e6ecd : SUCCESS in 8m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dcb84a8741c54119977f0be363fb2c5d : SUCCESS in 40m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e0c0ffc544734ec78c3bf4071cb76499 : FAILURE in 7m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cb467e4f0e26401ab50adf485afe3ae7 : SUCCESS in 1h 16m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8890bb712681440c9411715cf87061d2 : SUCCESS in 1h 36m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/561bf56fa1084f1fab7554adfd7ac27c : SUCCESS in 1h 41m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f22c6bd566c749ae95a1580066c952e0 : SUCCESS in 1h 08m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/db8a4e1fb1ed4169b6d78b8d719fe32a : FAILURE in 1h 20m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c353b38e8516497bae7e98abd25c3bce : SUCCESS in 41m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5eab3ccbfca54208a86a7e01768dc231 : SUCCESS in 1h 35m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/02c95f93f4a1460ba51c88b0df025adf : SUCCESS in 2h 30m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7135ba70b6c345b79f705d27558cbe63 : SUCCESS in 55m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c257926f3704bbaba757c92c959e247 : SUCCESS in 42m 47s","accounts_in_message":[],"_revision_number":3},{"id":"84511da10aded4f80eda0177f5b45749cdf81fb5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-10-28 14:33:17.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4f1b645/ : FAILURE in 2h 15m 05s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/815582/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/043cf45/ : TIMED_OUT in 3h 03m 11s","accounts_in_message":[],"_revision_number":3},{"id":"ca3b4f6ca64491efbe6e9a9af399f5c7bb0a3029","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-28 14:46:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/815582/3/check/iscsi-driver-nimble/c5a97d7/c5a97d7.html : SUCCESS in 1h 48m 33s","accounts_in_message":[],"_revision_number":3},{"id":"d02f94b8da58fd55751c9c724f0f5bf7313b3107","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-10-28 14:59:15.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/815582/3/check/tempest-dsvm-full-focal-py3/9bd02b7/ : FAILURE in 1h 27m 12s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/815582/3/check/devstack-plugin-ceph-dsvm-focal-py3/8295002/ : FAILURE in 1h 35m 49s","accounts_in_message":[],"_revision_number":3},{"id":"c6d5d60b519b436b69042466048a6367b55e00ae","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-28 15:23:52.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/3/check/DellEMC_Unity_FC/267ceb6/DellEMC_Unity_FC/120 : FAILURE in 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/3/check/DellEMC_Unity_iSCSI/76ce061/DellEMC_Unity_iSCSI/114 : FAILURE in 21m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"dc14a1b188c74a49ddbf8510688497e4d922f402","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-28 17:44:13.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-815582-3/ns5_nfs : SUCCESS in 4h 39m 38s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-815582-3/ns5_iscsi : SUCCESS in 4h 40m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"a2251db3cd31ec88fbd02b88ea3013a7b57b0e37","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2021-10-28 17:47:15.000000000","message":"Patch Set 3:\n\nBuild Successful. ibm-gpfs-dsvm-tempest-full: https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/74/index.html","accounts_in_message":[],"_revision_number":3},{"id":"ea9994f72afebaf7accc3428244f3408cb9663fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2021-10-29 00:27:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ee349768aa48c74733443e6271bbe0576de361b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2021-10-29 00:29:58.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"792e46d12d7999d4c640e2b9d9aa5e93af94456e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 00:53:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f6e573c0d59400aae048519314a26e1 : SUCCESS in 15m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9d5f0685c6ff4486b3c818d609222e67 : SUCCESS in 22m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"16c04f905cd276cf8dfb42f32f8dac46aa1814f4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-29 01:22:36.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/82/815582/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/434bee1/434bee1.html : FAILURE in 50m 55s","accounts_in_message":[],"_revision_number":5},{"id":"a9303abd3b42c264dd2f9b9057b56a7639bcb96f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-29 01:35:36.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/815582/5/check/iscsi-driver-alletra-6k/1bf525c/1bf525c.html : FAILURE in 1h 04m 23s","accounts_in_message":[],"_revision_number":5},{"id":"c126b6274ba9fb0517e81024d2236d7d4c891ec0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-10-29 01:50:16.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/82/815582/5/check/dsvm-tempest-cisco-zonemanager-job/39a0e91/ : SUCCESS in 1h 12m 15s","accounts_in_message":[],"_revision_number":5},{"id":"9d7cc14a64b5f2866ea481145aa5f8ba9c97d644","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-10-29 01:51:46.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/5/iscsi/ : FAILURE in 1h 20m 30s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/5/smb/ : SUCCESS in 1h 09m 01s","accounts_in_message":[],"_revision_number":5},{"id":"3d83d476bf224ec9d4eb1d964473ee1c4c2e9629","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-29 01:52:20.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/82/815582/5/check/iscsi-driver-primera/3f09ccf/3f09ccf.html : SUCCESS in 1h 20m 39s","accounts_in_message":[],"_revision_number":5},{"id":"8df8d200ff077b95498871069adad9b45d272076","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-29 02:24:00.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/815582/5/check/iscsi-driver-nimble/772cda4/772cda4.html : SUCCESS in 1h 49m 48s","accounts_in_message":[],"_revision_number":5},{"id":"e6965946b3e22b12042b007386317204d4ddba31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 03:10:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba7863c96e9547a9baad9a86fe435eba : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1145ced4590443858e8b65c6cbbc5579 : SUCCESS in 9m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88bd9dac7c324e16ac8c0c88727ef5f6 : SUCCESS in 11m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e403e4191a394170a79297eb3c148948 : SUCCESS in 11m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a9544fc17c74fbc80af3597b39b659c : FAILURE in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/a3ffd183162e4151922a99bcb677e3a4 : SUCCESS in 55m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3838455c87874d3ca7f309934f3adeb5 : SUCCESS in 1h 05m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cf8f452240b2465fbb9e7c2b93a060c7 : SUCCESS in 18m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dc430eb43ce1434c88735917cf29cb97 : SUCCESS in 6m 28s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/421de27af90a432c85da8a0369bdbd49 : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0315c5f23d304d9d86eecb0ce8f84226 : SUCCESS in 8m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ac322f12e25348c2b3478309957bdcfa : SUCCESS in 8m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/774f53817df145a48b3308e8e5a455e0 : SUCCESS in 37m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5914814d2ce94c8ba4da20beb8b82d1a : FAILURE in 4m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aa5a873b5ab14af4b6d88a3bba19aa54 : SUCCESS in 1h 16m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/05ed41d9da7f462db79477ab0d99d104 : SUCCESS in 1h 39m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d93320a085d342c29c3e61e3d27a70d7 : SUCCESS in 1h 37m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0b41eb27f7bc440ca965fced6cafe27b : SUCCESS in 1h 03m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/569c96350ae14a2585cfaf7157d2fdaf : SUCCESS in 1h 13m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/515121f0b33d48cb8ff4ba4f8b5b9bbd : SUCCESS in 47m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/920c25f62d394d70ab14b3206a67d6e1 : SUCCESS in 1h 06m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/94917d990abd4514a138d3795b4255c0 : SUCCESS in 2h 31m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f73ff87d893042058fc8408b580e31ed : SUCCESS in 47m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2dc82c0fba39462ab8bc5e7633376af2 : SUCCESS in 44m 46s","accounts_in_message":[],"_revision_number":5},{"id":"9260ecbfb0810b3202a8ab6cd563141297f23510","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-10-29 03:36:23.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/062f99b/ : SUCCESS in 2h 06m 53s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/815582/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0582ace/ : TIMED_OUT in 3h 05m 22s","accounts_in_message":[],"_revision_number":5},{"id":"871bd8e650b00c053706a3520d8f8e53c94ea04d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-29 03:40:56.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/5/check/DellEMC_Unity_FC/f06c5d8/DellEMC_Unity_FC/129 : FAILURE in 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/5/check/DellEMC_Unity_iSCSI/9b99929/DellEMC_Unity_iSCSI/123 : ABORTED in 3h 10m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"dfb2c53836f278d5ac08258ec2573a5cda3f0208","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-10-29 03:43:53.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/815582/5/check/tempest-dsvm-full-focal-py3/3e3a5e9/ : FAILURE in 1h 25m 11s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/815582/5/check/devstack-plugin-ceph-dsvm-focal-py3/e59329f/ : FAILURE in 1h 30m 52s","accounts_in_message":[],"_revision_number":5},{"id":"4bb3e9c6de3965dc91bb8663c6872dffc54554fd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2021-10-29 04:01:01.000000000","message":"Patch Set 5:\n\nBuild Successful. ibm-gpfs-dsvm-tempest-full: https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/76/index.html","accounts_in_message":[],"_revision_number":5},{"id":"2e8c6b6bf7a1f63373b7746f8a35fb729a9050bb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-29 04:44:03.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-815582-4/ns5_nfs : SUCCESS in 4h 7m 38s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-815582-4/ns5_iscsi : SUCCESS in 4h 16m 18s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"23f057bc6fe42cfcfe4863e6b0c34f20f2edf44f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-29 05:32:12.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-815582-5/ns5_nfs : SUCCESS in 4h 13m 55s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-815582-5/ns5_iscsi : SUCCESS in 4h 19m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"2febf1c602a3e6c4fafed6afc5d94da08a1e65e4","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-10-30 13:02:17.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/5/20211030185738/NEC-ISCSI : SUCCESS in 2h 47m 52s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/5/20211030185738/NEC-FC : SUCCESS in 3h 04m 35s","accounts_in_message":[],"_revision_number":5},{"id":"9311c94d0498b1543e9531f3c1e9d4c03b1b0f99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2021-11-25 00:53:13.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"fb4784b74429da017452c08657e6ef060cea9241","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 01:17:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/57551507e4cd4a328fb145797d5bd379 : SUCCESS in 21m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad6c402b9348461db7675d858a8a4b0b : SUCCESS in 16m 13s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6d2a9eef26ffb5657fb3a78d40e985d9d379a59b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-25 01:49:19.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/6/iscsi/ : SUCCESS in 55m 13s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/6/smb/ : SUCCESS in 52m 48s","accounts_in_message":[],"_revision_number":6},{"id":"c5e09ff5746f7bb36ce587ef252917ce770af4a8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-25 02:08:00.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5556d5f/ : SUCCESS in 1h 13m 12s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/815582/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/934bd98/ : SUCCESS in 1h 00m 16s","accounts_in_message":[],"_revision_number":6},{"id":"c9d3e26290f118eb791e2473bfc3f3e0afbc5dc2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-25 02:11:15.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/82/815582/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/74d03d4/74d03d4.html : SUCCESS in 1h 15m 31s","accounts_in_message":[],"_revision_number":6},{"id":"a80c3dcc5969d0325cc99f4a6454c4f8b4d6d408","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-25 02:13:29.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/82/815582/6/check/iscsi-driver-primera/42ca0cb/42ca0cb.html : SUCCESS in 1h 17m 30s","accounts_in_message":[],"_revision_number":6},{"id":"64014d157d888dac4c3e3d7efa5ab69f12313097","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-25 02:45:02.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/82/815582/6/check/dsvm-tempest-cisco-zonemanager-job/c8eeaed/ : SUCCESS in 1h 08m 55s","accounts_in_message":[],"_revision_number":6},{"id":"0b53d1eab6b00ce9a65b0d9054e248b1ab7c01fe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-11-25 02:49:11.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/6/check/DellEMC_Unity_FC/dd71967/DellEMC_Unity_FC/191 : FAILURE in 1h 06m 23s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/6/check/DellEMC_Unity_iSCSI/5e64052/DellEMC_Unity_iSCSI/193 : FAILURE in 22m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"7bf2aa6f1ee352e0543aa4057d09e993ee836c94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 03:18:38.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12e3f73eaa7d440ea3b8f2dd8d081d57 : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d24fc48f6844a5fb488e9b1871af493 : SUCCESS in 9m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e79e34801830454e9bfd7abef25b77aa : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/264e67a2abeb48f7b4aaa0f32eb7b263 : SUCCESS in 12m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/349e339d66f04494918327c0879b9c43 : SUCCESS in 53m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e810039fe807425bb62c87afedd6cc02 : TIMED_OUT in 2h 12m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ce8509d0d6b449259f8ac3d282da2e1b : SUCCESS in 12m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d01e669f9e4447118ca3889f7353ccf8 : SUCCESS in 6m 08s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f71cd7d2ec8a4678a87e134db9856448 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/95b34c8200064d6085c6fea3bc29b8c4 : SUCCESS in 7m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/194f99d5b6534a51abf6cad5e38294ed : SUCCESS in 6m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b12e4190ef7f4d6d95456aa9dd616ba3 : SUCCESS in 40m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e1caa0f8dd28473b9767e3c93e753e27 : SUCCESS in 5m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/68fae69e4c2d42cf806bdf76bc15820d : FAILURE in 1h 03m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7e9fc598c2a44512a043f1e0dcf79fb8 : SUCCESS in 1h 30m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4af1b9e769314515abe43b1fd1bbc4b8 : SUCCESS in 1h 20m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e27f2be5a7c842e7bbaa46b441687e33 : SUCCESS in 1h 00m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9be30e330c394ee195772686d4e138bd : SUCCESS in 1h 12m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/83f88450950b4413987eadd9af852a0f : SUCCESS in 45m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/706aaa119e0d4aebace52387fe72929f : SUCCESS in 1h 11m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b05c99a8464c455786592d56ab525d81 : SUCCESS in 2h 18m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6dfc1ee6eba74309a82e6e1232c9d27d : SUCCESS in 43m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bfbe19b647f4bd49b6314265aeb294a : SUCCESS in 40m 09s","accounts_in_message":[],"_revision_number":6},{"id":"acc41569bf0a9e7d9052966e2c772f6c14a10c4e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-25 03:43:58.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/815582/6/check/tempest-dsvm-full-focal-py3/8b368e6/ : SUCCESS in 2h 09m 22s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/815582/6/check/devstack-plugin-ceph-dsvm-focal-py3/9b4723a/ : SUCCESS in 2h 30m 08s","accounts_in_message":[],"_revision_number":6},{"id":"d560ae437af2eff4cd4add2e8533d9e2b844f645","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2021-11-25 04:07:46.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7dd1f85159c5a4ad813afad23fbead74d1d935fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 04:33:49.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c18c37a26a2e4725be5c774f1de8bcae : SUCCESS in 18m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/605d94b05e1c4017a400a6b375140b60 : SUCCESS in 23m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1db2ca86ce1de45a5f3580199ac96fc980d74fce","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-25 05:01:49.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/6/iscsi/ : SUCCESS in 53m 25s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/6/smb/ : SUCCESS in 51m 26s","accounts_in_message":[],"_revision_number":6},{"id":"37e812589131cc3a056ecb45a45652cbdc7062a9","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-11-25 05:15:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/0c457d8/index.html : SUCCESS in 1h 06m 46s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1b489a6/index.html : SUCCESS in 1h 07m 24s","accounts_in_message":[],"_revision_number":6},{"id":"bdaddbdbd96ee752fc6e3aada132a8692ffa2af0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-25 05:23:21.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/82/815582/6/check/dsvm-tempest-cisco-zonemanager-job/caf8c03/ : SUCCESS in 1h 09m 27s","accounts_in_message":[],"_revision_number":6},{"id":"0c77f0787f5e816a8c199f2280d694de8cae07bd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-25 05:30:00.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/815582/6/check/iscsi-driver-nimble/664013d/664013d.html : FAILURE in 43m 34s","accounts_in_message":[],"_revision_number":6},{"id":"123b821643722a4781cf381b808c1b236de50b40","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-25 06:04:07.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/82/815582/6/check/DellEMC_XtremIO_FC/e1c891d/DellEMC_XtremIO_FC/232 : SUCCESS in 1h 10m 42s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/6/check/DellEMC_XtremIO_iSCSI/e16a7dd/DellEMC_XtremIO_iSCSI/249 : SUCCESS in 1h 48m 14s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"e153ae8bece364361f1228203281efc8ccd45742","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-25 06:21:11.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/815582/6/check/tempest-dsvm-full-focal-py3/59a8d16/ : SUCCESS in 2h 08m 32s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/815582/6/check/devstack-plugin-ceph-dsvm-focal-py3/ede3b3f/ : SUCCESS in 2h 08m 54s","accounts_in_message":[],"_revision_number":6},{"id":"0c9fb523d3225a0e7c46e0426aab24d5e0fbd3bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 06:31:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dac3bbc2d454c1fabb3fa5cf5e33d47 : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/266ea69cd70043b586d82e0ab152d761 : SUCCESS in 11m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9efc93efed948eb9a735731b5e554a3 : SUCCESS in 19m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b541699324fc4d938af5c530ac03bb74 : SUCCESS in 11m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/f56c88a7f8ad46cbaed227938a60344e : SUCCESS in 1h 12m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ec49aab67aff497ea449c0f3516befe5 : SUCCESS in 1h 11m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8e01b3530c2d47a4baf82c194f9965e3 : SUCCESS in 13m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/87536d969874412c9e5d74248c031934 : SUCCESS in 5m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d20956a2c3a84276bf5c26fd1f11f623 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d6b48f74615b4a7abdf495e6a4a88fec : SUCCESS in 8m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4fd3007b7c18404585242acc5b99c352 : SUCCESS in 6m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/13a086cca7a744b0b89e84f034058067 : SUCCESS in 44m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/97244a0d6ad449c0b4419fb37545b23c : SUCCESS in 7m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/604762f7e6c04af8a9db5ffe9856567b : SUCCESS in 1h 37m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f210c2ad21c9489f90528f5f39ad6745 : SUCCESS in 1h 30m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/430a1dbb4f494a27b429979ac8ae22e1 : SUCCESS in 1h 26m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/37e600f67f7a44efb5c4e64d05f4536c : SUCCESS in 1h 09m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/20b52851eddf41baa96aa8d37e35896a : SUCCESS in 1h 13m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a2e70ee83f714a2981c11fcb9f21194f : SUCCESS in 35m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d9e2a751f877477e97a3bb81322f3327 : SUCCESS in 1h 04m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ba11583000c40eb8f5dde8e523b1658 : SUCCESS in 2h 15m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbfcf0a9ec4546be9cad4df7c4845ec2 : SUCCESS in 46m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1196b5a6006400c8d927a75d7b8e67a : SUCCESS in 47m 45s","accounts_in_message":[],"_revision_number":6},{"id":"d0070530ca6d355b4ccf78d68e04312a3e0f111b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-11-25 07:43:17.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/6/check/DellEMC_Unity_FC/d714dd7/DellEMC_Unity_FC/192 : FAILURE in 1h 11m 52s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/6/check/DellEMC_Unity_iSCSI/32e83f2/DellEMC_Unity_iSCSI/194 : FAILURE in 24m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"b2b4e4f89d0ab68259abf6b95d5c8acd52a3d2ac","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-25 08:19:35.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/82/815582/6/check/DellEMC_XtremIO_FC/a062488/DellEMC_XtremIO_FC/233 : SUCCESS in 1h 12m 49s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/6/check/DellEMC_XtremIO_iSCSI/94e0b21/DellEMC_XtremIO_iSCSI/250 : SUCCESS in 1h 48m 07s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"f8bd052ccb0d369c3d642a71023c432a578bda6f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-11-25 08:24:45.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/82/815582/6/check/DellEMC_SC_FC/7e99889/DellEMC_SC_FC/67 : FAILURE in 27m 18s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/6/check/DellEMC_SC_iSCSI/55d268f/DellEMC_SC_iSCSI/66 : SUCCESS in 1h 53m 24s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"96d1075d1fa3a51d36a1dd67c0944318b2e069f6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-11-25 08:24:50.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/82/815582/6/check/DellEMC_PowerFlex/f338235/DellEMC_PowerFlex/147 : FAILURE in 1h 53m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"10739005cb71d9283b7e8da8f7303c8bb3eafe69","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-25 08:51:00.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/82/815582/6/upstream-check/cinder-cDOT-iSCSI/ad17503/ : FAILURE in 1h 58m 56s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/82/815582/6/upstream-check/cinder-cDOT-NFS/8c2a4f5/ : SUCCESS in 1h 22m 13s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/82/815582/6/upstream-check/cinder-cDOT-FCP/e21198d/ : FAILURE in 2h 18m 33s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"99d43f9282c817835640ef6f9ecd61b92b102c3f","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-11-25 08:58:49.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/815582-6/HPMSA-iSCSI : SUCCESS in 1:12:50\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/815582-6/HPMSA-FC : SUCCESS in 44:50.17","accounts_in_message":[],"_revision_number":6},{"id":"9645318091b8d404a9f77609ce3914da44c7fc0a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-11-25 09:31:55.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/6/check/DellEMC_VNX_iSCSI/63f0330/DellEMC_VNX_iSCSI/133 : FAILURE in 3h 00m 23s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/82/815582/6/check/DellEMC_VNX_FC/4b67bed/DellEMC_VNX_FC/134 : FAILURE in 25m 16s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"a245defaa0437fa372ec7982d70a0eb69c65c1f9","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-11-25 10:21:57.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/815582-6/Lenovo-iSCSI : SUCCESS in 1:28:25\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/815582-6/Lenovo-FC : SUCCESS in 57:05.76","accounts_in_message":[],"_revision_number":6},{"id":"73eeb79b4a2b2afabfb0fda86d992896f7ef0cb5","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-11-25 10:25:16.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/82/815582/6/upstream-solidfire-check/cinder-solidfire-iSCSI/07d0c7e/ : SUCCESS in 1h 35m 01s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"9a5e248d68c63d236e110ba51ce9cfa6fd9d10ff","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-11-25 17:12:29.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/82/815582/6/check/DellEMC_PowerMAX_FC/9c0033a/DellEMC_PowerMAX_FC/75 : SUCCESS in 3h 38m 37s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/6/check/DellEMC_PowerMAX_iSCSI/9c34721/DellEMC_PowerMAX_iSCSI/73 : FAILURE in 3h 26m 04s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"6f25260d4ca8fac89de6b10c3d8d83c5cc939d20","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2021-11-26 09:56:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ae2aafd476570184b52c72a352d3deae4babece4","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-11-26 17:32:41.000000000","message":"Patch Set 6:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/82/815582/6/check/dsvm-tempest-TOYOU-ACS5000-driver/652005c : SUCCESS in 2h 11m 06s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/82/815582/6/check/fc-test/adbe62f : SUCCESS in 2h 09m 37s","accounts_in_message":[],"_revision_number":6},{"id":"cfc87ff720c900c362f946dff5dd5f075beabd8d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-27 04:17:27.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-815582-6/ns5_nfs : SUCCESS in 4h 41m 51s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-815582-6/ns5_iscsi : SUCCESS in 4h 46m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"76cbadbe9905f404544bdb64b72075caac7ee68f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-27 09:13:40.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/6/20211127152402/NEC-ISCSI : SUCCESS in 2h 41m 38s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/6/20211127152402/NEC-FC : SUCCESS in 2h 49m 35s","accounts_in_message":[],"_revision_number":6},{"id":"4cfb889eeb82d36e52c44e34e7596f217a4b88a3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-12-06 07:44:16.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/815582/815582/6/check-not-vote/ext-cinder-zuul/634d8a3 : FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"6474b6da196da95b9b2b9a05552bc87648a89d10","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2022-01-12 15:24:53.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"834c250c7f3abee027999078328e9f1ecb811c2c","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-01-13 10:43:59.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"27a50873c9d36336e0fcdce576045e1e1c981245","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-01-17 08:47:44.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"3dc63d0280b300b8ddc9b7787ed90a269e783413","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-17 08:59:50.000000000","message":"Patch Set 7:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/17_01_2022_08:53:09\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"e14d9bb4cfb6c6a5a6014415fd499028a7a35c8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 09:13:18.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c139014703984d1fa73d78d74d6fd968 : SUCCESS in 22m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0487f5809bb4ab5b8cace92a23987e7 : SUCCESS in 21m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a9e2ceb418c264c9025ed91b7ad39ad6c1a1ba94","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-17 09:32:10.000000000","message":"Patch Set 7:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":7},{"id":"6e02e4e480944a376d15479263194339e062b2f8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-17 09:59:11.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/7/iscsi/ : SUCCESS in 54m 15s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/7/smb/ : SUCCESS in 1h 10m 45s","accounts_in_message":[],"_revision_number":7},{"id":"7c8c96230c3857715385f6250c8ac0a27907785f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-17 10:13:39.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/815582/7/check/tempest-dsvm-full-focal-py3/3d39948/ : FAILURE in 1h 05m 54s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/815582/7/check/devstack-plugin-ceph-dsvm-focal-py3/70feb6b/ : FAILURE in 1h 12m 22s","accounts_in_message":[],"_revision_number":7},{"id":"cebf20dbf7306b76c9967802a07ad8e7eddb0bbb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-01-17 10:56:15.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/28dda74/ : SUCCESS in 2h 04m 52s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/815582/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/8329bca/ : SUCCESS in 1h 03m 31s","accounts_in_message":[],"_revision_number":7},{"id":"ae95b547d91eb3c3ed9d706074f82a436303b7db","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-17 11:25:01.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/7/check/DellEMC_Unity_FC/6f9151b/DellEMC_Unity_FC/401 : FAILURE in 1h 28m 11s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/7/check/DellEMC_Unity_iSCSI/84c7c20/DellEMC_Unity_iSCSI/389 : SUCCESS in 1h 24m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"0f492ae3d7052820f07831e1f5183ee19ab6681c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 11:25:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35419db961594c898f9e12ba376b59e0 : SUCCESS in 8m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a3312cc3ab545f2b2193c30a9f0605c : SUCCESS in 10m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03489efe0e174b009008f04cfda202e1 : SUCCESS in 11m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5a785ceceb6479cbaa4858a6fa51a88 : SUCCESS in 12m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/7546aacec09c498b8b56fa2892916506 : SUCCESS in 1h 20m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ce16780fdca14b1aa2264de455a4d82a : SUCCESS in 1h 07m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c0c1f31bf9743c6b0ed02c6c6acd766 : SUCCESS in 48m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0552592a4eac415ab5e35e2f01be0383 : SUCCESS in 12m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3cc7346b9f5b4aaab3826b5fea124158 : FAILURE in 7m 16s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d368213ed9284602acf1056320632b62 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/87c54ebb58e64d01bebad06ff8c71373 : SUCCESS in 7m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0a08ef870d89482eaf7ce2222e32a59a : SUCCESS in 9m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/000dc0a211204b928fb6a9addf8c698c : SUCCESS in 43m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1ae8ad3f72e94e869ca40c9141e3f143 : SUCCESS in 7m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bebb3bfbb66d49d1bf9a8cef744dccc8 : SUCCESS in 1h 18m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0e29092aaea44435a987638fb656fd72 : SUCCESS in 1h 23m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/70d6be125f0f496291d61f1bdc1b5a1f : SUCCESS in 1h 33m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/be1d33bb36a840cfaa398dc4fed60698 : SUCCESS in 53m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e41c03db3ac84b6696f70e8a6aca739a : SUCCESS in 1h 23m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e61fb57911dc41ca81bc40c2b6aa4574 : SUCCESS in 41m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/263419f6cc6e4f31a48371733bac249c : SUCCESS in 1h 17m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4214849bb38345159d860db05381dd16 : SUCCESS in 2h 27m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fe29f5b9c70458c88106c700fcaca51 : SUCCESS in 51m 32s","accounts_in_message":[],"_revision_number":7},{"id":"762758c9d3ed5e05ff06fecd2bdf55b5e616caf6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-17 11:25:59.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/815582/815582/7/check-not-vote/ext-cinder-zuul/b81dcba : FAILURE in 1m 04s","accounts_in_message":[],"_revision_number":7},{"id":"d6280e5bd64a681605ec1a5ad0bcd3554deba112","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-01-17 11:35:30.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/82/815582/7/upstream-solidfire-check/cinder-solidfire-iSCSI/c26d4ce/ : FAILURE in 7m 32s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"e15cb3513f09da4d045539387f8c6e866ca9ff0f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-01-17 11:42:30.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7acc921dc5a448a299e43f87df9842f9 : FAILURE in 1h 30m 37s","accounts_in_message":[],"_revision_number":7},{"id":"0a591d18154625df4a1101d30e1e7d4a1296c64a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-17 12:14:52.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/82/815582/7/check/DellEMC_PowerFlex/71d3f1f/DellEMC_PowerFlex/266 : FAILURE in 48m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"d218426e01da887238b3d2f32a13168419d33e52","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-01-17 12:20:53.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/815582/7/ : SUCCESS in 1h 36m 00s","accounts_in_message":[],"_revision_number":7},{"id":"81b20827f4c1258de8c7187cee0950ecd37fd7df","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-01-17 12:22:06.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"287c684bdbfdeddcc71ec1caedb55afc8a6721d2","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-01-17 12:29:00.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/82/815582/7/check/DellEMC_XtremIO_FC/19a1ac9/DellEMC_XtremIO_FC/478 : SUCCESS in 56m 36s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/7/check/DellEMC_XtremIO_iSCSI/597fb18/DellEMC_XtremIO_iSCSI/491 : SUCCESS in 1h 23m 56s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"8009096385d555e21bd532baf6de2932982b1e43","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-01-17 12:52:33.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-82-815582-7/1642418753 : SUCCESS in 37m 33s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":7},{"id":"1d49d7993b73ef2e6c9151b7d8b162db1cc44e59","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-17 12:52:53.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/82/815582/7/upstream-check/cinder-cDOT-iSCSI/496fa35/ : SUCCESS in 34m 37s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/82/815582/7/upstream-check/cinder-cDOT-NFS/46d94b8/ : SUCCESS in 39m 56s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/82/815582/7/upstream-check/cinder-cDOT-FCP/880e1ed/ : FAILURE in 1h 25m 26s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"9eb1024b0c41ee4597d3f8ecf101ccab09050856","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-17 14:42:37.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-815582-7/ns5_nfs : SUCCESS in 4h 26m 20s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-815582-7/ns5_iscsi : SUCCESS in 4h 33m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"c76d9e8c8e803eaf680ecce48bec7c3b2355ff74","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-01-17 15:14:32.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/7/check/DellEMC_PowerStore_iSCSI/b3d60f6/DellEMC_PowerStore_iSCSI/236 : SUCCESS in 1h 28m 33s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/82/815582/7/check/DellEMC_PowerStore_FC/4dcadb6/DellEMC_PowerStore_FC/240 : SUCCESS in 1h 11m 45s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"c044aa7c0efba50c44c55014082206541476a539","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-01-17 16:15:59.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/82/815582/7/check/DellEMC_SC_FC/af3d749/DellEMC_SC_FC/204 : SUCCESS in 49m 54s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/7/check/DellEMC_SC_iSCSI/18c33b2/DellEMC_SC_iSCSI/201 : SUCCESS in 1h 52m 20s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"e25324c1bded6ac907e2b644cef6b0f13e500a1e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-01-17 17:27:18.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/7/20220117232753/NEC-ISCSI : SUCCESS in 2h 48m 31s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/7/20220117232753/NEC-FC : SUCCESS in 2h 59m 22s","accounts_in_message":[],"_revision_number":7},{"id":"a54430deef08d84884aeb4896206fc4b51639c17","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-01-17 19:31:35.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/7/check/DellEMC_VNX_iSCSI/e815339/DellEMC_VNX_iSCSI/201 : FAILURE in 2h 53m 22s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/82/815582/7/check/DellEMC_VNX_FC/f45dcee/DellEMC_VNX_FC/211 : SUCCESS in 2h 41m 27s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"011f212c39e8a2f21c37c6f5d79cecb613f150a7","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-01-17 19:39:46.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/82/815582/7/check/DellEMC_PowerMAX_FC/bc4545f/DellEMC_PowerMAX_FC/181 : FAILURE in 1h 23m 22s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/7/check/DellEMC_PowerMAX_iSCSI/2534b93/DellEMC_PowerMAX_iSCSI/194 : FAILURE in 1h 14m 19s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":7},{"id":"e22c929a627ceadf55e7ccc766ccbd9b5d8bcd4a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-17 20:04:07.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/7/check/DellEMC_Unity_FC/57e120d/DellEMC_Unity_FC/407 : FAILURE in 1h 16m 28s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/7/check/DellEMC_Unity_iSCSI/7d7f129/DellEMC_Unity_iSCSI/395 : SUCCESS in 1h 21m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"82a7eb4dd3f5b2943f41b076efe369d1d160ef52","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2022-01-18 05:59:05.000000000","message":"Patch Set 7:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/82/815582/7/check/dsvm-tempest-TOYOU-ACS5000-driver/856f00d : SUCCESS in 1h 49m 38s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/82/815582/7/check/fc-test/7ac67ae : SUCCESS in 2h 09m 31s","accounts_in_message":[],"_revision_number":7},{"id":"a7be952769e89446fed5623b62863f9dfb42adc1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-01-20 14:07:22.000000000","message":"Patch Set 7: Review-Priority+1","accounts_in_message":[],"_revision_number":7},{"id":"96490c15856727b5498dc1f653e7ec1dc836db9f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-01-20 21:19:15.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"dca411387560692f5776d318c19d4e5ff61907f6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-01-20 21:22:23.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1b5c18b538e74ebba7a79b5b878c8259ff0dda73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-01-21 08:09:42.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"cdae230164feab0dbe408c5a0f2ea2e6ade41972","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-21 08:09:51.000000000","message":"Patch Set 8:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15012/","accounts_in_message":[],"_revision_number":8},{"id":"e9b8025e8a97bbe13650c4b19a7c466a487b7636","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-21 08:11:47.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15012/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"c12d03d2aa96b75142765999f8a14b2fb5bc7c73","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-01-21 08:15:26.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2d0535975f2bb41f6a8944b7c1642cc9925913c8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-21 08:17:28.000000000","message":"Patch Set 8:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/21_01_2022_08:09:51\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"4f924721f539fff8a398ef71cc7cb8888cb7ca59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 08:32:25.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e541d1204d214fbd83fccfdb0250483d : SUCCESS in 18m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bbcc33b768f84ea4ae29ee2edb9cf003 : SUCCESS in 19m 27s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"81ffcb766f1bc2faceae71a753b35ef3a848e091","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-21 08:59:33.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/815582/815582/8/check-not-vote/ext-cinder-zuul/8df5ac8 : SUCCESS in 35m 37s","accounts_in_message":[],"_revision_number":8},{"id":"2ba8c1e8d91c97c8dd25767aa1f6481310cfc31a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-21 09:08:00.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/815582/8/check/tempest-dsvm-full-focal-py3/8965ad9/ : FAILURE in 52m 30s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/82/815582/8/check/devstack-plugin-ceph-dsvm-focal-py3/39a90ef/ : FAILURE in 52m 29s","accounts_in_message":[],"_revision_number":8},{"id":"e8420d9e7dac61eb28309ae6b74529a721f926a3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-01-21 09:11:09.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/815582/8/check/iscsi-driver-alletra-6k/5196651/5196651.html : FAILURE in 59m 55s","accounts_in_message":[],"_revision_number":8},{"id":"e66b71077be2a6827a57bf97d07db83b28bb4065","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-21 09:15:24.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/8/iscsi/ : SUCCESS in 56m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/8/smb/ : SUCCESS in 1h 04m 59s","accounts_in_message":[],"_revision_number":8},{"id":"f64800029e1f8ff449af7cab62e1ea3a1a554e6a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-01-21 09:27:48.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/82/815582/8/check/dsvm-tempest-cisco-zonemanager-job/bcbf91b/ : SUCCESS in 1h 05m 28s","accounts_in_message":[],"_revision_number":8},{"id":"8c426cf62cfa16956a062bee6acadb9378a0a98c","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-01-21 09:29:18.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/82/815582/8/check/DellEMC_XtremIO_FC/e6b309a/DellEMC_XtremIO_FC/527 : SUCCESS in 55m 40s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/8/check/DellEMC_XtremIO_iSCSI/7ef4582/DellEMC_XtremIO_iSCSI/537 : SUCCESS in 1h 18m 50s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"dda55c247512da617f6426393facc38dd85f3eb7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-21 09:43:23.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/8/check/DellEMC_Unity_FC/8815143/DellEMC_Unity_FC/443 : SUCCESS in 1h 04m 27s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/8/check/DellEMC_Unity_iSCSI/329cfcb/DellEMC_Unity_iSCSI/435 : SUCCESS in 1h 33m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"d1f431a7db99fa4c39ae3c0f99bacb7547af3117","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-01-21 09:44:34.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/af1b2682876f48edb53331b213fc3535 : FAILURE in 1h 32m 01s","accounts_in_message":[],"_revision_number":8},{"id":"57bbf76f6ca76b68d0de2e63daec31c93bc2f3a2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-01-21 09:47:56.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/e317f83/ : SUCCESS in 1h 36m 17s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/815582/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6bd8656/ : FAILURE in 1h 18m 04s","accounts_in_message":[],"_revision_number":8},{"id":"bbd9970d8d7c54710148fc33a92d140ffe0e1647","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-01-21 09:52:12.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1047629ec12893949652d563799eca97183832cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 10:27:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5af8a1c9abea41abaf87988e4a4b051d : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f5882b09a3e49189c0244e0503e6fcf : SUCCESS in 10m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7746e6f1689d421c8dc59a84d0ee0776 : SUCCESS in 10m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79251bdc5439432caec4dfe67cd2b02d : SUCCESS in 13m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/40747386be6448b1a524a172c27e3d34 : SUCCESS in 54m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c12e3242301f4c009b17e4f724403d7c : SUCCESS in 1h 15m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de263f01da964fea8a2ccf0cb83ab436 : SUCCESS in 48m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a69f75c1ac404cc8b1d6f95480e20ad7 : SUCCESS in 14m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5058842b6a1e46099541bc6f2b584380 : FAILURE in 6m 52s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b52dd69f6ced43d4b14aa51fa729981b : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3286b7aa92b45179b3edc65deb10e7e : SUCCESS in 7m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/58ae5498b6fb4a0789e866a545ffeb8f : SUCCESS in 7m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4c1878f4a2234c8d834d65ae76cc5a7b : SUCCESS in 42m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/be0ab3b9e1f2425ba910cb7e86233359 : SUCCESS in 7m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c70efb0996c44b90943e50a83444714c : SUCCESS in 1h 21m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fb8e5919322f4672a90b58148ea61e9b : SUCCESS in 1h 25m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/828a791d86874420a9caea2b12993118 : SUCCESS in 1h 38m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e16295cf417942559e1886a99cdccc74 : SUCCESS in 1h 12m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5118f3d744824af8b3eb10f3d9b06e9b : SUCCESS in 1h 14m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a755957433b841abaeef47f4e0d45763 : SUCCESS in 42m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5558d57cb7d24d7b95093024be66a62d : SUCCESS in 1h 20m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/51d72e4f965f4233b2bbb88c717de081 : SUCCESS in 2h 12m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23539e6698e54ecc9642526c8a647b93 : SUCCESS in 47m 08s","accounts_in_message":[],"_revision_number":8},{"id":"152f5a8576bf4e6749c8f9e6b211234628dd000a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-21 10:33:23.000000000","message":"Patch Set 8:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/21_01_2022_10:27:26\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"9861a6df4a34e0a17fbf6d8acb34f5d1fd834e58","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-01-21 11:05:07.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/82/815582/8/check/DellEMC_PowerMAX_FC/2574f04/DellEMC_PowerMAX_FC/206 : FAILURE in 37m 14s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/8/check/DellEMC_PowerMAX_iSCSI/c11ff25/DellEMC_PowerMAX_iSCSI/219 : FAILURE in 34m 31s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"5da70ba5a110699be8cb6339e3cb34582f34d2ce","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-21 11:05:55.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/82/815582/8/check/DellEMC_PowerFlex/d10c9e4/DellEMC_PowerFlex/293 : FAILURE in 37m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"702e8a695b8459798887ba2b92df0965c50cb76e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-01-21 11:57:47.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/8/check/DellEMC_PowerStore_iSCSI/2d26000/DellEMC_PowerStore_iSCSI/262 : SUCCESS in 1h 30m 04s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/82/815582/8/check/DellEMC_PowerStore_FC/9563d06/DellEMC_PowerStore_FC/265 : SUCCESS in 1h 07m 46s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"063122683f4820c51aaebed6c75fee012fea43c3","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-01-21 12:01:54.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/82/815582/8/upstream-solidfire-check/cinder-solidfire-iSCSI/57cf840/ : FAILURE in 1h 32m 42s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"532abf7e45a6388ce2b811afd49438eb63987b61","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-21 12:05:44.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/82/815582/8/upstream-check/cinder-cDOT-iSCSI/3870963/ : SUCCESS in 33m 44s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/82/815582/8/upstream-check/cinder-cDOT-NFS/5b49c31/ : FAILURE in 1h 37m 04s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/82/815582/8/upstream-check/cinder-cDOT-FCP/fb2c9fc/ : SUCCESS in 1h 19m 45s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"9a1d098b7e89b99ec178a3d761d5e5db4dd2c89a","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-01-21 12:10:32.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/82/815582/8/check/DellEMC_SC_FC/00d84d8/DellEMC_SC_FC/229 : SUCCESS in 55m 50s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/8/check/DellEMC_SC_iSCSI/46fb957/DellEMC_SC_iSCSI/227 : FAILURE in 1h 42m 55s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":8},{"id":"48df7dfdcb95a4a29e386c5cae728bef218578da","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-01-21 12:31:06.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/815582-8/HPMSA-iSCSI/ : SUCCESS in 49:26.82\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/815582-8/HPMSA-FC/ : SUCCESS in 43:55.74","accounts_in_message":[],"_revision_number":8},{"id":"e312dbec01bbc94ce9fff6e2d857eabfff490c76","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-21 12:46:22.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-815582-8/ns5_nfs : SUCCESS in 4h 35m 42s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-815582-8/ns5_iscsi : FAILURE in 4h 30m 47s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"fddf14c5ae57f5dfca3ffaf7c17ae0837e38463e","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-01-21 13:00:19.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/815582-8/Lenovo-iSCSI/ : SUCCESS in 1:04:09\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/815582-8/Lenovo-FC/ : SUCCESS in 58:01.61","accounts_in_message":[],"_revision_number":8},{"id":"373901b90664c3deee55e063593ac15457cb3551","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-01-21 13:03:38.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/815582-8/PVME-iSCSI/ : SUCCESS in 1:05:30\n- PVME-FC https://os-logs.seagate.com/dell-ci/815582-8/PVME-FC/ : SUCCESS in 1:01:46","accounts_in_message":[],"_revision_number":8},{"id":"72d6101de8f64a7f658bb2eb2e0c8b6ce635c58e","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-01-21 13:07:39.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/815582-8/Seagate-iSCSI/ : SUCCESS in 1:05:28\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/815582-8/Seagate-FC/ : SUCCESS in 1:03:40","accounts_in_message":[],"_revision_number":8},{"id":"f5504e4308f2bba8d5fccc96d97d80c739f314c5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-01-21 13:34:15.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/8/20220121213157/NEC-V-ISCSI : FAILURE in 54m 22s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/8/20220121213157/NEC-V-FC : FAILURE in 1h 02m 14s","accounts_in_message":[],"_revision_number":8},{"id":"07e7a9d74c485efa245a811c010750043f67af78","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-21 14:15:36.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/8/check/DellEMC_Unity_FC/a8b28cc/DellEMC_Unity_FC/447 : SUCCESS in 1h 06m 47s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/8/check/DellEMC_Unity_iSCSI/35d1cb9/DellEMC_Unity_iSCSI/439 : FAILURE in 12m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"1010934312701cf5bd981ef111758db195a51221","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-21 14:31:27.000000000","message":"Patch Set 8:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/21_01_2022_14:26:44\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"739d092a111faa079a3d81180beb98bf9d52d27b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-01-21 15:12:29.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/82/815582/8/check/DellEMC_XtremIO_FC/11ae3f0/DellEMC_XtremIO_FC/531 : SUCCESS in 54m 54s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/8/check/DellEMC_XtremIO_iSCSI/fd42fc8/DellEMC_XtremIO_iSCSI/541 : SUCCESS in 1h 20m 32s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"f5b7411662b21a8f1b5a0e49a0c8f13baa620ca3","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-01-22 07:08:26.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/815582/8/ : RETRY_LIMIT in 2m 09s","accounts_in_message":[],"_revision_number":8},{"id":"75beac4ce25efb8dce40a10984f56b0d314ee849","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2022-01-25 14:20:24.000000000","message":"Patch Set 8:\n\nBuild Successful. ibm-gpfs-dsvm-tempest-full: https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/174/index.html","accounts_in_message":[],"_revision_number":8},{"id":"3047f22d5dee06fd5a547f536e3661530e02d653","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-01-25 22:18:28.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/8/20220126035334/NEC-FC : SUCCESS in 3h 10m 09s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/8/20220126035334/NEC-ISCSI : SUCCESS in 3h 24m 51s","accounts_in_message":[],"_revision_number":8},{"id":"2061926764eb91b4f52847143cba7d4397c3f8aa","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-04-27 07:32:37.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2f0838c1629fcff427078eaad106dcbedd8dbe4d","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-05-18 01:58:13.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7e6c204df27369fea8b3de004a7c73a8345d01d5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-06-02 12:24:52.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fb6a84ee534dca8ac1a2ae7282c20874eb8775c5","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-06-06 00:29:53.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"233fd15ddd1a724c55c7ab8db8c4a66a3ce11248","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2022-06-17 12:23:55.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3f398a8771a356bae3a0b21ea3a816692b789a1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-06-23 00:46:26.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"54b9287a00fc53345b1477c304da4aba44a988f1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-23 00:48:57.000000000","message":"Patch Set 9:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1655945194817 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"8dfebc30c5c7187f812ebac7ac48629f9e0e3d5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 01:07:16.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03526131f08e4252a285497e993f0c25 : SUCCESS in 15m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c171103ef52b4345a93b8b70a769f908 : SUCCESS in 16m 01s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"385b0775ec9ac5a0d1ea53aeb68030ce949d2740","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-23 01:29:07.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/dfe793b36e374850b55447abdc3b8df8 : SUCCESS in 41m 32s","accounts_in_message":[],"_revision_number":9},{"id":"857a93f4bcdec41a02d9bd07aa74e8fcb457e27e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-23 02:11:54.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/82/815582/9/check/dsvm-tempest-cisco-zonemanager-job/2e937e6/ : SUCCESS in 1h 19m 49s","accounts_in_message":[],"_revision_number":9},{"id":"4b8a0a6a988d1dcfa513adb168c19ffedbbac6f7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-23 02:20:49.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/9/iscsi/ : SUCCESS in 1h 21m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/9/smb/ : SUCCESS in 1h 32m 42s","accounts_in_message":[],"_revision_number":9},{"id":"4550f4a024705a93dbe4aad74e5108d47b7bc805","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-23 02:26:40.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k-focal https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/815582/9/check/iscsi-driver-alletra-6k-focal/01e7a83/01e7a83.html : FAILURE in 1h 31m 18s","accounts_in_message":[],"_revision_number":9},{"id":"b069c6fcf962c1c4780090621b7c38d1a5198b53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 02:51:51.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cab3b4803dc42b9a60416b6d303772e : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13466624813a414cb18b96127959dad6 : SUCCESS in 11m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e657ebf1b464e3f8b481dbb25aa6102 : SUCCESS in 9m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7330f768ad54ebc97039d2f59b916d7 : SUCCESS in 11m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83faa54e7e5a4cc4bc8c6cd4d66b169b : FAILURE in 7m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/1207b877437240d489f65935a27c8b20 : SUCCESS in 54m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c326f54ecaf4e76bbb39005b572ebf9 : SUCCESS in 1h 07m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6738d56b0a3a477d98f2582881724834 : SUCCESS in 44m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7663aef2574c4defb1fa31fbb542bdbf : SUCCESS in 6m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/24cdaa207cfa4e29a368c5b7cc41d328 : SUCCESS in 14m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cf3ef49bb9de40e986752b38a9f05331 : SUCCESS in 4m 52s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6f6c5ae6c0204af2926e151e6ecec007 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d407b4bd27db4d4a828ee7b7b6f0781c : SUCCESS in 5m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cdae7cf3c7094c61a48adace8ff443d9 : SUCCESS in 7m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/16fbbcd9e9154c0199e443720f659674 : SUCCESS in 29m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/feb2f94e90e04ccbaf1c0adb7653178c : FAILURE in 6m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/852e9c6ca1254bc9aee88ab50edba4c5 : SUCCESS in 1h 06m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f4a8c87645a44a16867edc642e085e05 : FAILURE in 1h 21m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0a116ca05a344d009fbba71c8616cf59 : SUCCESS in 1h 28m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1c544209425b49bf9f0b4350757ba51f : SUCCESS in 1h 03m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f3bbe37f6644453d89e66b3c2f696432 : SUCCESS in 1h 10m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b08591cafc8e4c1dbd900c693b352b83 : SUCCESS in 35m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c207e41f950843a2bdf02b591a778f4c : FAILURE in 59m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/21ed3f55493d460aa41bc07f8a958001 : SUCCESS in 2h 03m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50c916d4a54f4818beebe90aea8e97ce : SUCCESS in 45m 07s","accounts_in_message":[],"_revision_number":9},{"id":"a92de70892591aa7e700d39188eecf580f14e548","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-23 03:37:30.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/815582/9/check/iscsi-driver-nimble/72ec436/72ec436.html : FAILURE in 1h 33m 04s","accounts_in_message":[],"_revision_number":9},{"id":"fd71800d9d7dac62b6490295bde8f512ba40f3d4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-23 03:51:13.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/96fce95/ : TIMED_OUT in 3h 02m 46s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/815582/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b589175/ : FAILURE in 2h 25m 46s","accounts_in_message":[],"_revision_number":9},{"id":"f558b849ca0dbd52a7c92a905e38c202d93f3c46","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-23 04:14:42.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-815582-9/ns5_nfs : SUCCESS in 3h 27m 47s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-815582-9/ns5_iscsi : SUCCESS in 3h 17m 22s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"e3ad00539d3cf4a0ccbbf7fa89eccb4d0293a948","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-23 04:23:21.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/b5b9c887e20849908b53396c08b71db8 : POST_FAILURE in 2h 38m 00s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/acc013ac0352423da044af271072eef3 : TIMED_OUT in 3h 05m 26s","accounts_in_message":[],"_revision_number":9},{"id":"8c383b35075df166e4ab6d5d616724565f84a417","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-23 08:01:55.000000000","message":"Patch Set 9:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":9},{"id":"76281b23610c5d16f3809c808947efc749e0c645","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-06-23 08:23:02.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"8b677cd4566ddf284025682b0c8af2079855d940","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-23 08:52:14.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/10/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/795e9fe/ : FAILURE in 28m 05s","accounts_in_message":[],"_revision_number":10},{"id":"3c7b0c6d9105952433ae5ca5566d46e1c7204ab6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 09:01:04.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5804a2c19c444cbf8860cf209ef63aa4 : SUCCESS in 14m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dbc8acb855cc40bc9a35a1dae8326e1b : FAILURE in 25m 43s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0a6538571763deda652cb521c7c7ed6aaa92cb99","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-23 09:05:57.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9f7996c8cef240758a69318dc6e67abf : SUCCESS in 41m 40s","accounts_in_message":[],"_revision_number":10},{"id":"b3b503c9aadd9f454d13439c3c861373dd877e5d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-23 09:19:14.000000000","message":"Patch Set 10:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16141/","accounts_in_message":[],"_revision_number":10},{"id":"29fc2f03eb62fa06b6807b730183908ad7d7a6e5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-23 09:21:22.000000000","message":"Patch Set 10:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16141/ : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"420fde2cb4efa0178df05d6ad9201e758ec40a74","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-06-23 09:34:35.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b432d8e3f6814facc75b47940e032da31de9f03f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-23 09:49:01.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/10/iscsi/ : SUCCESS in 1h 01m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/10/smb/ : SUCCESS in 1h 25m 23s","accounts_in_message":[],"_revision_number":10},{"id":"63bbaa7695d352f18184ee7a63c92ebb0a1d9c11","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-23 09:55:08.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k-focal https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/815582/10/check/iscsi-driver-alletra-6k-focal/0baf443/0baf443.html : FAILURE in 1h 30m 39s","accounts_in_message":[],"_revision_number":10},{"id":"aedf0856dcfab0244e0b3f7615c9ba55893e5073","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-23 10:11:37.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/82/815582/10/check/dsvm-tempest-cisco-zonemanager-job/a6825a5/ : SUCCESS in 1h 10m 15s","accounts_in_message":[],"_revision_number":10},{"id":"5ae7a159a80ca482395622a782a1f4951b5a2255","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-23 10:33:55.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/33975f730aca470e9cffae90fdc2febe : POST_FAILURE in 1h 56m 09s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/0bf77e88f8c648a583142cd865061266 : POST_FAILURE in 2h 07m 27s","accounts_in_message":[],"_revision_number":10},{"id":"fb0c9b3677a750c19bee1ba52caba676edcbaec7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-23 10:46:33.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/815582/10/check/iscsi-driver-nimble/377055b/377055b.html : FAILURE in 1h 33m 40s","accounts_in_message":[],"_revision_number":10},{"id":"03503ea004fda81c1778f1047163fb9657a8e38c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 10:49:02.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95bce013431644049608cb30890ce968 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d3ee1beed3d416d8cac1106e559320e : SUCCESS in 11m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/294150bce07e47c3a2162fa46dc5fdcc : SUCCESS in 11m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df238fb2ab364c7baedfef8d63f38ae7 : SUCCESS in 10m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c422df072fe44eb68ce040cd591801b5 : SUCCESS in 12m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/b28bfdbbfbd446a18186429f7775808a : SUCCESS in 59m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/17f320f55db645a1a3e445fb92a42487 : SUCCESS in 1h 25m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/570eae1aaeb34a9aaa52281a8247676e : SUCCESS in 50m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/546bc268a5a841019bd637e65f6a4ae2 : SUCCESS in 11m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cd7b615648f14c9a91f3b4ab7acaeee9 : SUCCESS in 14m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/58e2e8a838674e2b8ad286b5640068b4 : SUCCESS in 5m 54s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d214b8ec4d2c428e80a23d2814057c35 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2a043be7a3754a4b82919dfcd79b3feb : SUCCESS in 6m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/00bbb054734b441aa6d5c8144f41c391 : SUCCESS in 7m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5853533e2ef64ea0aec4187c9fc15724 : SUCCESS in 40m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4c2ef679eb024c969d234201470e1c6c : FAILURE in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b6dec4e1c1684f0a995d56bbf8e8398f : FAILURE in 1h 11m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7a2a12f015524758bd07f7d9c133e367 : FAILURE in 1h 26m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6ded0eaf76a14c41a054e60e3c076885 : SUCCESS in 1h 31m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/62dc6c77573f497c9058f23bdbc800ae : SUCCESS in 1h 11m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/368f93a7db034a48930d4861cfd69656 : SUCCESS in 1h 18m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/219914071228415cb16d4d7fcb1f8685 : SUCCESS in 34m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c0a038b01d904468a8997436743c00eb : FAILURE in 1h 11m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e008de8f059f43cbb0c47375ffa82d3d : SUCCESS in 2h 20m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c11f5bd5a5a4b8985dbaa29a079023e : SUCCESS in 53m 08s","accounts_in_message":[],"_revision_number":10},{"id":"e9650b48e55f762e196913951faa7073c9528351","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-23 11:28:27.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/10/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4375e95/ : TIMED_OUT in 3h 03m 32s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/815582/10/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/23ee67f/ : FAILURE in 1h 12m 31s","accounts_in_message":[],"_revision_number":10},{"id":"7ea802713a8777e09af3fe5dd4d699287371e40b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-23 12:02:49.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-815582-10/ns5_nfs : SUCCESS in 3h 23m 18s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-815582-10/ns5_iscsi : SUCCESS in 3h 39m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"74c62ef078f3d5238c6f32d521d99817303d8902","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-06-23 12:53:51.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4582227f5c9871ebd280e12c2516807e61901bdb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 13:19:11.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/374c583a1d724943b98b6c378dd104e4 : SUCCESS in 15m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc8af812e46a409dae80a477aae12de8 : SUCCESS in 23m 25s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"511f07de2f4a087aa1dc2f4c533ecaeb35a608e1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-23 13:23:33.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/10/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d2a795c/ : FAILURE in 28m 21s","accounts_in_message":[],"_revision_number":10},{"id":"20e592d1adfd5d3dd22384019a25d16e46d85ef6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-23 13:57:30.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/10/iscsi/ : SUCCESS in 57m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/10/smb/ : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":10},{"id":"1d45ed8fab4b8248c9f7c1f2b811c1dc06063599","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-06-23 14:19:11.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi finger://vm-css-bionic-2/666fda5afd874565b925c03b1ca72936 : POST_FAILURE in 1h 16m 35s\n- storwize-tempest-dsvm-full-fc finger://vm-css-bionic-2/0a915d738e3144fbad0a127f8ea3f94f : POST_FAILURE in 1h 24m 28s","accounts_in_message":[],"_revision_number":10},{"id":"6c6f1ca880bc34c312e7dc8392dca673b70f9a75","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-23 15:08:00.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/acac8f2c40fd451a9f8194409ced1247 : POST_FAILURE in 1h 33m 45s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/9887e1dbc1f14f38a6a4e6d67432d2d6 : POST_FAILURE in 2h 11m 24s","accounts_in_message":[],"_revision_number":10},{"id":"b5112c7ad9d33e758923a95fd79cc017010e6987","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 15:12:34.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbda95d7b0c047d089f76d0538c41c6b : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/034830f9ef7f4f2d98775a7a66248662 : SUCCESS in 11m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d56f5dafbf7f45e391808237cbe29e48 : SUCCESS in 11m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa988c67ecbf4b87a703062a8e5d87c6 : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abc556136a2f469184096eb5f5dd6b22 : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/8e7d0d4ce65f4d4aa3a73d5bae316d21 : SUCCESS in 52m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/95aed84371d44aba9c4d1dbb6c252f39 : SUCCESS in 1h 07m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae07608f602a420caaa07ee34cea54ef : SUCCESS in 46m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d82fa3280bc461c945d75af0f285002 : SUCCESS in 6m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ce7077767ab04318a58963ea100025b1 : SUCCESS in 14m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1bdb1cb6b6cf45bea050418010b8f127 : SUCCESS in 5m 54s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/95416f293d5c4a1fb47d9e737e3bbf9d : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f5e5daffed2b417f97339f5388deae3e : SUCCESS in 6m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b5b0ec8f440b41f8a8e44a8371b414da : SUCCESS in 6m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0debb0cf151c4fd4ab6d04944c6c369a : SUCCESS in 49m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d05148e06ab84a57ae553cb2069bdd3c : FAILURE in 6m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c36362e656ca43bc8aa5952e2087ea0c : SUCCESS in 1h 12m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1dbddecdd240480a9b8e884f523a1a44 : FAILURE in 1h 30m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/30de3152f46d4f5f863f078a4c5bbf2a : SUCCESS in 1h 36m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6907fa536728436697c9e54844c61a5d : SUCCESS in 1h 04m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/05944cca87c94410ac04643b1c28a5c4 : SUCCESS in 1h 13m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1339f3d29fc048e09b53aeedf592365f : SUCCESS in 40m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c2d17f927c794fe491276e91fda04dba : FAILURE in 1h 00m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ee620edbfc89428c943dec3c689e259d : SUCCESS in 2h 13m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4192086583204820b1f1187d9999d9b9 : SUCCESS in 47m 49s","accounts_in_message":[],"_revision_number":10},{"id":"2dcd92af86ff19066b38726c89b5847f09cda146","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-06-23 15:29:56.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-82-815582-10/1655997154 : SUCCESS in 17m 19s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":10},{"id":"16031da7460958cd5a9e0114b46c1c3d15770028","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-23 15:34:46.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e219452987f54f37920275f559fff307 : SUCCESS in 46m 13s","accounts_in_message":[],"_revision_number":10},{"id":"fa680d919c7aff576706cdfa08e67dfd66ca1aea","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-23 16:33:56.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/10/upstream-check/cinder-tempest-plugin-ontap-fcp/b69987d/ : SUCCESS in 1h 03m 10s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/10/upstream-check/cinder-tempest-plugin-ontap-iscsi/16a9ffd/ : FAILURE in 48m 42s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/10/upstream-check/cinder-tempest-plugin-ontap-nfs/9e38cc8/ : FAILURE in 1h 19m 55s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"8237c36fd1198af2da8cab8697b04d3564ec9c2d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-06-23 16:43:44.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/82/815582/10/check/DellEMC_PowerFlex/2688c79/DellEMC_PowerFlex/896 : SUCCESS in 1h 30m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"2bd6579ff33a2c4ba5d13004ec37b8309c23ce2e","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-06-23 17:29:25.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/82/815582/10/check/DellEMC_SC_FC/eabf180/DellEMC_SC_FC/742 : SUCCESS in 47m 50s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/10/check/DellEMC_SC_iSCSI/4697593/DellEMC_SC_iSCSI/764 : SUCCESS in 1h 54m 24s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":10},{"id":"aa51fb1bbea96b33c570bb27af4d1f9788f22419","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-06-23 18:03:06.000000000","message":"Patch Set 10:\n\nBuild failed.","accounts_in_message":[],"_revision_number":10},{"id":"0700d9fc25e710811b78c639a4f370d34ffd21c7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-23 20:36:44.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/82/815582/10/check/dsvm-tempest-cisco-zonemanager-job/9a888b8/ : SUCCESS in 1h 10m 43s","accounts_in_message":[],"_revision_number":10},{"id":"6e3043e91f7714e222242761044b3b6d856d96f1","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-06-23 20:53:00.000000000","message":"Patch Set 10:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/815582-10/HPMSA-iSCSI/ : FAILURE in 1:00:15\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/815582-10/HPMSA-FC/ : FAILURE in 57:12.17","accounts_in_message":[],"_revision_number":10},{"id":"b1e4b490d8ba6d3cda0c2590790f6fad3b71756c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-24 01:57:36.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/10/check/DellEMC_Unity_FC/129e63e/DellEMC_Unity_FC/1009 : SUCCESS in 1h 14m 43s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/10/check/DellEMC_Unity_iSCSI/1739a79/DellEMC_Unity_iSCSI/947 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"5e5b676514f4de37f3ab08fa936c9410c532286c","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-06-24 02:39:14.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"06315af6370d66e0a7b72c726c45f0d502c21360","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-06-24 05:35:06.000000000","message":"Patch Set 10:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/815582-10/Lenovo-iSCSI/ : FAILURE in 1:09:38\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/815582-10/Lenovo-FC/ : FAILURE in 1:07:40","accounts_in_message":[],"_revision_number":10},{"id":"663734d738ef914469dad38379935bc803bf75db","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-06-24 08:22:00.000000000","message":"Patch Set 10:\n\nBuild failed.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/815582-10/Seagate-iSCSI/ : FAILURE in 1:10:26\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/815582-10/Seagate-FC/ : FAILURE in 1:08:28","accounts_in_message":[],"_revision_number":10},{"id":"5f641ddfda6fad29cbe6e0a1cab16c0c83d1109d","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-06-25 19:45:20.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/82/815582/10/check/DellEMC_PowerMAX_FC/6567758/DellEMC_PowerMAX_FC/616 : SUCCESS in 4h 06m 46s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/10/check/DellEMC_PowerMAX_iSCSI/f6cc6b4/DellEMC_PowerMAX_iSCSI/603 : ABORTED in 10h 30m 01s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":10},{"id":"40de3b32a7512b851952bb415fdf932b8cf6c804","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-07-13 01:41:39.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"dd781f9a8e66f9a05a7c2190d31b6644f1fa4318","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-07-13 01:54:43.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"509ce7615ffadb03e2aa96ee198b85f4c2b0ce59","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-07-13 02:05:14.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"26bc31b9ec0766b26efca9d1aed7b7a2f14da169","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-07-13 09:45:31.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c77fbd745ae311eb5ccf22ea567ef44531e4d00e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-14 20:37:45.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/10/20220715030131/NEC-V-FC : SUCCESS in 1h 59m 12s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/10/20220715030131/NEC-V-ISCSI : SUCCESS in 2h 36m 11s","accounts_in_message":[],"_revision_number":10},{"id":"88a06adaaa72c645acd57ed4a8970c5cdf423e9b","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-07-15 02:19:58.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b9b3f7ab8bddba5b654d41bf41237e2cf67623f1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-07-18 17:03:43.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/10/20220718192638/NEC-ISCSI : SUCCESS in 3h 28m 54s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/10/20220718192638/NEC-FC : SUCCESS in 6h 37m 02s","accounts_in_message":[],"_revision_number":10},{"id":"ef47486a783e8bb61a37e3c9f05a47adf4774f4b","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2022-07-18 22:56:05.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4813c51cc617a5edc5fe1693c7e264eaaad65747","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-07-26 12:12:59.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"42ded5c3e1fbc1631bb66cf9d928409fca30b7f5","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-07-26 12:19:56.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8c778268467e307fed123f9a67f2ea5fc61b68bf","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-07-27 01:18:05.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"efecc7c2958e061766a35a85e262b032220f83f8","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2022-07-27 06:31:19.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"326c18690aa18fac479de41233c9cbf5529a528f","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-07-27 07:21:34.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"cd8e8ab80b3f464697c961dea29af1771fa1c79d","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2022-07-27 21:11:26.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"72512f80a4b675033e14e5c95a5c519f70888690","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-07-28 00:18:36.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"6ccd24ffb826504879045627081b22eabd704fdb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-28 00:21:48.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":11},{"id":"90e857158b584e326c2a7bdac949d9e4f0f6a023","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-07-28 00:22:11.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"fe34391ce3f781869685f6a8dbb9023428001a50","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-28 00:37:55.000000000","message":"Patch Set 11:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1658967524875 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"f210526ac90c0440fa89c572e47f11299919b271","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 00:40:40.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c1520c41ea74178bdfa58a54cee258a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf8fd870a2b84a8ab294582e8754e083 : SUCCESS in 20m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5527884f5af84dafbaa45ee96afe7a10 : SUCCESS in 21m 28s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"307023d7293e37611ac8fd66cacdcac25e1e1963","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-28 00:48:29.000000000","message":"Patch Set 11:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/306","accounts_in_message":[],"_revision_number":11},{"id":"e1f02dd97ccf0cde8a550ccc42a97cf6696237d3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-28 01:22:11.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/815582/11/check/iscsi-driver-alletra-6k/5c16517/5c16517.html : FAILURE in 1h 02m 18s","accounts_in_message":[],"_revision_number":11},{"id":"9f311ff14b3298686dd46e20cd31717eec967e68","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-28 01:25:30.000000000","message":"Patch Set 11:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":11},{"id":"4563e3d66b0bc8683a892b5f62e8cbbd8b542fdd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-28 01:27:36.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/11/iscsi/ : SUCCESS in 1h 08m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/11/smb/ : SUCCESS in 58m 29s","accounts_in_message":[],"_revision_number":11},{"id":"6e3435714d0076b171770f7a17f2268244e590ca","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-28 01:45:18.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/815582 : SUCCESS in 1h 24m 25s","accounts_in_message":[],"_revision_number":11},{"id":"7d970d3758f3f8045e86d8c13afb4165f3f6a804","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-28 01:58:00.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/11/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8f94904/ : SUCCESS in 1h 38m 27s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/815582/11/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ca4220b/ : SUCCESS in 1h 27m 08s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/82/815582/11/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/62110cf/ : FAILURE in 33m 31s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"709de48573c63630412463b48ceb9b78fdcb8069","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-28 02:00:27.000000000","message":"Patch Set 11:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/82/815582/11/check/fujitsu-eternus-dx-iscsi/90b8b98 : FAILURE in 1h 37m 39s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/82/815582/11/check/fujitsu-eternus-dx-fc/f19a4a1 : FAILURE in 1h 40m 59s","accounts_in_message":[],"_revision_number":11},{"id":"12efed536abbee95daf3a91dce3442788f96c89a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 02:18:30.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32e13fe9ecc64891b95257e3b58fa107\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb062c7688ce4323a066b4370cb8ab19 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1354af1658b4e90b2085a5b4fd99991 : SUCCESS in 10m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5a240c11ffd400897ff5642dc1b699c : SUCCESS in 12m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c7f839c927e4487b0cd910c6bfaa406 : SUCCESS in 12m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f18422b65714a3099fd096fb7a5dbec : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/2603d65b22864e90b907d65248c81647 : SUCCESS in 52m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8fa11659252b46aa89e0a4ab26d8154b : SUCCESS in 1h 42m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6e1d5f94e8149599056d4ab7939add0 : SUCCESS in 45m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed6bf345322e4896af49aaa1c3906c0a : SUCCESS in 8m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0c39e37d8d0f411db6eddc63591787db : SUCCESS in 15m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4c02e2a1d7dc4fd88a49a775f1b03c8a : SUCCESS in 5m 27s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2c1e8b14180b4b13988f2635fb20a872 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/93958a3bc44b481ca8263f4c630f7b91 : SUCCESS in 6m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4f9e330233d14795b84b4ee6dc72639d : SUCCESS in 6m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9a5aa43d2f794c94ab8bdf1904bb46cb : SUCCESS in 34m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/653b7a55dbc24f4d9a71bf1272267ff3 : FAILURE in 6m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9868ed0e75df42d0b6bcb1cb3154025b : SUCCESS in 1h 11m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/134b7fa8629b4bea8647527070a8c89b : FAILURE in 1h 16m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/40586cc6b06e4652b1603f2916b3c044 : SUCCESS in 1h 24m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f635c201e8d44e0f9c16f3beed8a4c73 : SUCCESS in 1h 02m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4c4ca920794a4fce8283f025de441598 : SUCCESS in 1h 22m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/191dbcb6fe65422ca1d4d886d85263d0 : FAILURE in 36m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9ab567ff66164db7b12a3deeb6fa04a3 : SUCCESS in 1h 25m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3004ad2cdf6a49e8a12c4b74dbc2127c : SUCCESS in 1h 57m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8ec5223b97a4beaaeb785c7ed5ad1c6 : SUCCESS in 47m 42s","accounts_in_message":[],"_revision_number":11},{"id":"9643cb10f7e6e3073a4655d3710bd2b4f5c3bae5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-07-28 02:27:51.000000000","message":"Patch Set 11:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/28_07_2022_02_18_38\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":11},{"id":"c541de92104c3fbd7b8ad8d1e1fe2b20be5444c7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-28 03:02:13.000000000","message":"Patch Set 11:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/dd2b8bfcc5fb4e0b8457248e475018c2 : POST_FAILURE in 1h 44m 30s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/f850dfcc48a64583a7667a75c99df86c : POST_FAILURE in 1h 51m 53s","accounts_in_message":[],"_revision_number":11},{"id":"4871e65ad81f2dff2d85aa097f516eed00183131","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-28 03:05:30.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/11/upstream-check/cinder-tempest-plugin-ontap-fcp/a65f603/ : RETRY_LIMIT in 8m 05s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/11/upstream-check/cinder-tempest-plugin-ontap-iscsi/a257c71/ : FAILURE in 45m 18s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/11/upstream-check/cinder-tempest-plugin-ontap-nfs/5dccefb/ : SUCCESS in 45m 22s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/11/upstream-check/cinder-tempest-plugin-solidfire-iscsi/0288481/ : RETRY_LIMIT in 7m 37s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"9ea2d744b2bcfb634e11dc9ccd86bd9ed98aeeb2","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-07-28 03:12:01.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/82/815582/11/check/DellEMC_PowerMAX_FC/58aef09/DellEMC_PowerMAX_FC/764 : FAILURE in 17m 56s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/11/check/DellEMC_PowerMAX_iSCSI/8616fca/DellEMC_PowerMAX_iSCSI/747 : FAILURE in 53m 10s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":11},{"id":"9f4e419f4c3e6736143b4e496450c88cb7d261b4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-28 03:49:10.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-815582-11/ns5_nfs : SUCCESS in 3h 29m 14s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-815582-11/ns5_iscsi : SUCCESS in 3h 30m 5s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"8d1207bdd749eef9bc31fdc41f2087c43f05a14b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-28 03:53:21.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/82/815582/11/check/DellEMC_PowerFlex/42572f9/DellEMC_PowerFlex/1091 : SUCCESS in 1h 34m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"ee043c82c6abd38579e1b04d130b5d5f274bf01e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-28 05:05:07.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/82/815582/11/check/DellEMC_PowerStore_NVMeTCP/f22ab51/DellEMC_PowerStore_NVMeTCP/47 : ABORTED in 7s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":11},{"id":"dc5a86004d1e71280bce7e3c7e0424ad0b0ab52d","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2022-07-28 05:56:25.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"4b62c55cb19b332d43b5364e51d734c009faf4ca","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-07-28 21:06:45.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/82/815582/11/check/DellEMC_SC_FC/3c9bd56/DellEMC_SC_FC/822 : SUCCESS in 51m 41s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/11/check/DellEMC_SC_iSCSI/814e585/DellEMC_SC_iSCSI/841 : FAILURE in 2h 00m 17s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":11},{"id":"bdf58440de33098c976dbdd5ab151b4a632e0938","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-29 05:11:38.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/11/check/DellEMC_Unity_FC/ebf6e0c/DellEMC_Unity_FC/1288 : FAILURE in 19m 31s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/11/check/DellEMC_Unity_iSCSI/0a9bcb8/DellEMC_Unity_iSCSI/1175 : SUCCESS in 1h 53m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"1c4cbd26d5d76cc1ae3241ee83113eb7a82dfc26","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-30 03:40:26.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/11/20220730100926/NEC-V-FC : SUCCESS in 2h 01m 07s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/11/20220730100926/NEC-V-ISCSI : SUCCESS in 2h 30m 57s","accounts_in_message":[],"_revision_number":11},{"id":"a6c8e12f9465786d17e4101ee44d0b54ab382905","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2022-07-30 03:45:00.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a9b4a447ee0cf6f9e17bd80350a0b617cb1aa4f6","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2022-08-03 04:49:27.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"9de2f97c61770307d7d29f3347a7f52a5317405a","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2022-08-03 12:38:43.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"0fac582ba89ee5633099a79c016b805a7cbbcd2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-08-05 00:51:52.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"008a357fc06d084a1c6de962f11adf1faa859fad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-05 00:52:03.000000000","message":"Patch Set 12:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1659660720622 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"37c48e4fdee56b6ac33660fb53906f0f265d29e5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-05 00:54:58.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":12},{"id":"44700afbda4fceaf22e828917b41e44a60cb115c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-08-05 01:07:17.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"9b7d0517cf920dcc801c028b14e4a4604059ea11","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-05 01:07:19.000000000","message":"Patch Set 12:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":12},{"id":"c8f50230f1e5449e4dfec1c7248af7a5d6d646aa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-05 01:07:27.000000000","message":"Patch Set 13:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1659661645646 : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"21e7905732bf274b61b5d8fd006973b4ec560494","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-05 01:10:28.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"c74cda1288f5ad3f635d041865adfad6641f403f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 01:26:14.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/817b31b16db544d39a11fd1df8c8768d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f79256b78ca47aba9ec46c4611cb611 : SUCCESS in 14m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d30dc4c1ee9c4b7fac07b62e9905ba74 : SUCCESS in 14m 16s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"6ebf4f1484304641ebcf63a30e1b07d46ae03838","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-05 02:00:21.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/815582/13/check/iscsi-driver-alletra-6k/4ed551a/4ed551a.html : FAILURE in 51m 32s","accounts_in_message":[],"_revision_number":13},{"id":"573243b0c3178f3a7590c52b3affe0d5d83d5402","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-05 02:49:09.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/13/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/66db31d/ : SUCCESS in 1h 34m 53s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/815582/13/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b5837aa/ : SUCCESS in 1h 24m 08s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/82/815582/13/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/5e73ac5/ : SUCCESS in 1h 35m 08s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"70025de950371e49484c976feb8e9c2b0bf2970b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-05 02:51:04.000000000","message":"Patch Set 13:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/82/815582/13/check/fujitsu-eternus-dx-iscsi/6f2a402 : FAILURE in 1h 39m 32s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/82/815582/13/check/fujitsu-eternus-dx-fc/3585f5f : SUCCESS in 1h 40m 58s","accounts_in_message":[],"_revision_number":13},{"id":"eb2f0b20f09ef27bf24421a61b19b8b5e61c75db","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-05 02:55:15.000000000","message":"Patch Set 13:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/903813981b0b4e5b8a929956c3d084b6 : POST_FAILURE in 1h 28m 48s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/7d71847fcab645358627101e83f9bd8d : POST_FAILURE in 1h 38m 17s","accounts_in_message":[],"_revision_number":13},{"id":"ad0a24cb2aacd8e4b4253c9a84f0d8b3000cb77e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-05 03:00:07.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/13/iscsi/ : SUCCESS in 1h 33m 00s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/13/smb/ : SUCCESS in 1h 52m 08s","accounts_in_message":[],"_revision_number":13},{"id":"25e27a4cd889ce4ba63903f9d2c3836c9b1894b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 03:06:08.000000000","message":"Patch Set 13: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b94352610db94502a4be8f07b725c881\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f085ce6118a4495591c12940efd8f762 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac98c52db6354be49e84d1455c99343f : SUCCESS in 11m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e688b8735c274258b85636eea52563da : SUCCESS in 12m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0aa604bc3d241bfbf7d7413c5fb960f : SUCCESS in 13m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c21117dee9a40549f2ae9a6c2008248 : FAILURE in 6m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/7c00d9ec7544439b9af0e400c0383daf : SUCCESS in 1h 07m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d74aeeecab174b9d9db9eb34f45aac31 : POST_FAILURE in 1h 52m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/191e87c01739477ea9a8c2ae03050cc2 : SUCCESS in 57m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07104ba93d264ed78038737a1279f059 : SUCCESS in 11m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d3eefe7015484f5f85dd49d3dce289e8 : SUCCESS in 16m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e2c6047386904a3999164229b7220766 : SUCCESS in 5m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9446e6debc8049dabb8d07a110cb3969 : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/06a1117857054da29a22b42d05d8c916 : SUCCESS in 6m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cafaa86d2d434d6daee18155b471a7ba : SUCCESS in 6m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3854e32494b14df69bd9199b951d70ec : SUCCESS in 39m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3943f1d108fa43c1acfee9f291b21062 : FAILURE in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/196b345459614a828edfce375a1f763a : SUCCESS in 1h 19m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f72a0bc77a2348529e692343a6823d54 : FAILURE in 1h 29m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8389a3dbf9b946b28bec182d73a9ee5f : SUCCESS in 1h 27m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2b0b67ad847a4288b246e1048428964f : SUCCESS in 1h 21m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0c04c6da05b84edcbe776fd00b169a90 : SUCCESS in 1h 17m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/854b146973594319a86a5a015dd7afad : SUCCESS in 34m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b6313790e3a8410dbe74d688a9d938f3 : SUCCESS in 1h 40m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f5fab74f737f4ef8914aaca65117e703 : SUCCESS in 1h 57m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16a9ad9818a647748a24381cde5e35d7 : SUCCESS in 54m 17s","accounts_in_message":[],"_revision_number":13},{"id":"4d535f24a5c5e4960fc61f051b872eee0b36e0e4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-05 03:19:44.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/815582/13/check/iscsi-driver-nimble/f38fdd6/f38fdd6.html : FAILURE in 1h 46m 29s","accounts_in_message":[],"_revision_number":13},{"id":"08f42aa9142bb854aeab7df97913f35ea2988c23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-08-05 04:21:38.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"5b967b59b709e67eec028a2733b0f7b2ab15ddb4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-05 04:21:39.000000000","message":"Patch Set 13:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":13},{"id":"fab4249801c723cb55ac0e5b99d98bd5d4498bec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-05 04:21:49.000000000","message":"Patch Set 14:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1659673305962 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"b52c4a89a3b85eaeb3f28c9f68e6738b75465a7e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-05 04:23:16.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-815582-12/ns5_nfs : SUCCESS in 3h 30m 55s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-815582-12/ns5_iscsi : SUCCESS in 3h 23m 20s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"929d924e5fde784c1de2fd49cd94cae5ce5a3246","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-05 04:23:18.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":14},{"id":"a5e4eccf10efe7219479022d3ab759810d9764d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 04:45:26.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32dc0ff78a1e49e6bb28933b1fa324b9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5695245c9f6a47bc97a93decd23ed018 : SUCCESS in 21m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3efb8036379e4223a678ecb3bc4a56a7 : SUCCESS in 21m 59s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"8bd1a5938efb8da5a5793a447fe6a24831e5f2c1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-05 04:45:52.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-815582-13/ns5_nfs : SUCCESS in 3h 29m 28s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-815582-13/ns5_iscsi : SUCCESS in 3h 30m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"ca1dc4a11956f1bfd6d9cb354703bc4ac3bab82c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-05 05:57:55.000000000","message":"Patch Set 14:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/becef0110fb04956a875e0b94b292ca4 : POST_FAILURE in 1h 19m 34s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/62212c6d35ce4b689cf2f71d04aa95ce : POST_FAILURE in 1h 33m 20s","accounts_in_message":[],"_revision_number":14},{"id":"c25c3d45c64c4036a4622fa8c5233e3b38803ccf","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-05 05:58:04.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/815582/14/check/iscsi-driver-alletra-6k/d442c70/d442c70.html : FAILURE in 1h 35m 01s","accounts_in_message":[],"_revision_number":14},{"id":"c10589f3617f0f10d53fe84baf547e9dd52f829b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-05 06:03:03.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/14/iscsi/ : FAILURE in 1h 40m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/14/smb/ : SUCCESS in 1h 04m 32s","accounts_in_message":[],"_revision_number":14},{"id":"31399bb5e72bf8214e4e9fd3d888a1987938a25f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-05 06:07:11.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/14/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/1569d06/ : SUCCESS in 1h 23m 36s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/815582/14/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/e4c1347/ : SUCCESS in 1h 44m 26s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/82/815582/14/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/f5c360d/ : SUCCESS in 1h 24m 43s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"9e4227f68e289c02e7800e97d3961ed9984a7235","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-05 06:09:45.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/815582/14/check/iscsi-driver-nimble/ec17225/ec17225.html : FAILURE in 1h 46m 39s","accounts_in_message":[],"_revision_number":14},{"id":"dce497e5ed6cc68a06410a93f87d8bde9168eb8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 06:25:24.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0da94164b1ca47e99d575fc1d8290129\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1a706a183174f6cb608643d2328b2a6 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2435d6f2d2c9486690d17a4d65c40ebe : SUCCESS in 9m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6310ef8a34ff438c8e9341a50943771b : SUCCESS in 10m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e806ec0016e94870b2415c1b9583af85 : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db415b3ebb940009914946954c199db : FAILURE in 6m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/febfc74762704a22b77685215028c0e4 : FAILURE in 24m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ad795b8f72ef42d7bcab2a58fb16cf43 : SUCCESS in 1h 06m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bd12b44c4ec470284b344c328a6ca92 : SUCCESS in 44m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa20a4aede1f4855ba8505a7e337de8e : SUCCESS in 7m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a413e554d6454ed49af5a5419ec85daa : SUCCESS in 18m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f0e13b06c9534229a5e84db1823d5c6e : SUCCESS in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/112f69e61a6a4dd69be0b03bc65fee36 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/212e10651a2040418b56ded2f77269a7 : SUCCESS in 6m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/24fe3bdac8a1453cbcd44ff7dcc1f499 : SUCCESS in 7m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/22b8a0bb48e14cc58d8bdb6443416e0a : SUCCESS in 46m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/40664d9ee9e54523822329d594737adc : FAILURE in 5m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dcf1c6928f594815b382905a96c9497e : SUCCESS in 1h 07m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8775d780227046cb9459dae9ed7d5732 : FAILURE in 1h 14m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6ea8b42f7f4d4927bdf33fa50b68bf49 : SUCCESS in 1h 25m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f03c2152b4694951b2f2efe921ca72eb : SUCCESS in 1h 16m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8d926233ffe9431a906334eca487fe41 : SUCCESS in 1h 16m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ca2b0e94576645e38daf1cfb016e8fbb : SUCCESS in 29m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/728d0da5e8694f8383e4292809db363d : SUCCESS in 1h 12m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fbef4da5255c405381517086318ea858 : SUCCESS in 1h 51m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab8033ec09e646ba8867b042597f3a6e : SUCCESS in 45m 11s","accounts_in_message":[],"_revision_number":14},{"id":"ed685787b3530a07993c7a83bb1d6afa8e85bc5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-08-05 07:16:11.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"b82a6e88ef214592eac05974e8d4ddc82fcf15dd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-05 07:16:13.000000000","message":"Patch Set 14:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":14},{"id":"14a0ba46951e212ee1e4dd26910e76fa7a3a5945","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-05 07:16:23.000000000","message":"Patch Set 15:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1659683781251 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"6991245ec3374a88df76efd3e1f8de259e17eda7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-05 07:17:39.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":15},{"id":"29d2222bb26c049114ac1a5326a50a91f32441d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 07:39:35.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0627af394ec4cb6aa15b56e8eb87e44\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38b8831c251b40118eab115e499fc910 : SUCCESS in 17m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5f9e61fc85694af1b69b0d4b88d7fbee : SUCCESS in 17m 17s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"baf162bc82f40ea01421301715fe5b63ce1c99e3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-05 07:49:07.000000000","message":"Patch Set 14:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-815582-14/ns5_nfs : SUCCESS in 3h 23m 19s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-815582-14/ns5_iscsi : SUCCESS in 3h 26m 59s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":14},{"id":"0aba1f48ffd5557c0380b059a50ec8a8ed7ca734","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-05 08:18:06.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/15/iscsi/ : SUCCESS in 1h 01m 12s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/15/smb/ : SUCCESS in 57m 26s","accounts_in_message":[],"_revision_number":15},{"id":"3b72e34b4d1207bb9e3b648c6aaace06a75c58bc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-05 08:51:09.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/15/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/3efbc06/ : SUCCESS in 1h 28m 44s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/815582/15/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/79c5812/ : SUCCESS in 1h 34m 00s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/82/815582/15/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/e8888c3/ : SUCCESS in 1h 25m 31s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"0899f64067bd0a299fb718f6f034e89eedfd6eec","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-05 08:52:00.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/815582 : SUCCESS in 1h 33m 16s","accounts_in_message":[],"_revision_number":15},{"id":"80f75babec28d167696abbb94686b5f97efd0811","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-05 08:55:20.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/815582/15/check/iscsi-driver-alletra-6k/b303bee/b303bee.html : FAILURE in 1h 34m 05s","accounts_in_message":[],"_revision_number":15},{"id":"fcfe32f36e5357ef1016442c0919d243b6a0bca1","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-05 09:02:18.000000000","message":"Patch Set 15:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/82/815582/15/check/fujitsu-eternus-dx-iscsi/c0e5394 : FAILURE in 1h 45m 09s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/82/815582/15/check/fujitsu-eternus-dx-fc/6aff0af : SUCCESS in 1h 39m 05s","accounts_in_message":[],"_revision_number":15},{"id":"cf2d6d7c190bc94d805dd7738b147915d518cb55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 09:23:14.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c9cbf6cb25b461cadd9059e885a0ed7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/802d102919cb4edc92d54142b890f0be : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0e114043e5a4359a7022b4c782d4846 : SUCCESS in 11m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c112257ddc6547b794a95e0ed7fa8034 : SUCCESS in 11m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf2bf5d9f80f4b159cc730c6ccd18d1a : SUCCESS in 13m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3731e8ee00624494a43f713464cebc13 : SUCCESS in 12m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/ea40c5f642414aa2b65d907900c6bd18 : SUCCESS in 57m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f33c3ac9970942afa488df732ba369bd : SUCCESS in 1h 06m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf36673084844324a0cf6de01829e611 : SUCCESS in 49m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/051b3644c295441287f73246178b3a9c : SUCCESS in 7m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f12cadcf4d9a49a1aabc3d671d98d1b0 : SUCCESS in 15m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e3cf47eed1d842fda0ed41220ce8b559 : SUCCESS in 8m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8f9cfbf5cabe4d7683e890e7cfb2d788 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6f136ce79c1848a78bfcccbe8bdf0f9b : SUCCESS in 8m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d5231720de1b4e33bd14393549993f27 : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a0d0cf6182624933b6c4015993ac02c9 : SUCCESS in 41m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2ae759736f8646b4a665fd9944153609 : FAILURE in 6m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c00b02508b6544da97235e49cabb77cc : FAILURE in 1h 16m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/729ef054ed8e4d2e8a9d4bf5b5f4b42f : FAILURE in 1h 35m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/448e1c256c974de59c14b4a4bd64d1fd : SUCCESS in 1h 39m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bf629a67a1e0402798be07bcecaef54a : SUCCESS in 55m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/08e93191964642d19658ca2e11cbeae3 : SUCCESS in 1h 26m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/517d2745bbad4e698596d16e729783d6 : SUCCESS in 43m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1d868ab1fada44759a7ef7c527741803 : SUCCESS in 1h 31m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/67385d2e5bf94a949f97aed375e689e3 : SUCCESS in 2h 05m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75ed765c916244e7aec395e1ea82b42e : SUCCESS in 1h 02m 47s","accounts_in_message":[],"_revision_number":15},{"id":"e34156b24367b9b20bc9624284ea9f31bc835e34","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-05 09:36:02.000000000","message":"Patch Set 15:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/2a4da7d079f7493abddadbab851bc9bc : POST_FAILURE in 1h 57m 05s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/333763457d614946a7addffe2d461b4f : POST_FAILURE in 2h 15m 39s","accounts_in_message":[],"_revision_number":15},{"id":"a152cab5487b2a3e87adde07617448f2372525f6","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-08-05 09:44:57.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"2ce44f83c7085f9cfbcd2f919c350fbd116c09d5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-05 09:46:39.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":15},{"id":"5b426cb9c6991960707caafc8421e8222dc56370","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 10:03:32.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2731853325814348ab5d75fbe43a3e4f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/801a47e859c742f7815fad5746769680 : SUCCESS in 13m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e81a824efcf34fe3b0c67350ed40a89f : SUCCESS in 13m 48s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"101ad82e1ec66c98c922f3d5a37b2431311e29ed","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-05 10:11:39.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/815582/15/check/iscsi-driver-nimble/66b2858/66b2858.html : FAILURE in 1h 46m 24s","accounts_in_message":[],"_revision_number":15},{"id":"06744c2bf75ae19c4b1e9360f33cec53c6dc7011","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-05 10:43:39.000000000","message":"Patch Set 15:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-815582-15/ns5_nfs : SUCCESS in 3h 26m 1s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-815582-15/ns5_iscsi : SUCCESS in 3h 26m 58s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":15},{"id":"5a9a29ddca67b3b48372f432f20fce664037a4e7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-05 10:48:27.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/15/iscsi/ : SUCCESS in 1h 02m 55s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/15/smb/ : SUCCESS in 50m 11s","accounts_in_message":[],"_revision_number":15},{"id":"4f395a084ec8513972a14a60a105e8c91425ad78","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-05 11:05:47.000000000","message":"Patch Set 15:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/2fe86a96ee2348158533d75940f44b44 : POST_FAILURE in 1h 12m 58s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/b8fc24a8548447cbb9f9ee7b892274f7 : POST_FAILURE in 1h 18m 29s","accounts_in_message":[],"_revision_number":15},{"id":"b82faa3d773d936806e9446a41d18da859a2c2e0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-05 11:27:01.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/15/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/01d7a5e/ : SUCCESS in 1h 39m 32s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/815582/15/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/81125d6/ : SUCCESS in 1h 26m 20s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/82/815582/15/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/a543027/ : SUCCESS in 1h 40m 59s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"969e9d5ef6a07c4fdc06e4ab0884715269d6525b","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-08-05 11:49:06.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi https://test-cinder99.s3.ap-south-1.amazonaws.com/storwize-tempest-dsvm-full-iscsi/a71000d/index.html : SUCCESS in 1h 38m 53s\n- storwize-tempest-dsvm-full-fc https://test-cinder99.s3.ap-south-1.amazonaws.com/storwize-tempest-dsvm-full-fc/7ccc6dd/index.html : TIMED_OUT in 2h 02m 55s","accounts_in_message":[],"_revision_number":15},{"id":"f5c759f8fbbf8b5ce4c1b1b5da7791091e7c5aa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 11:50:13.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/219983057f7446c6ad6ba806474e65f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8d4e59d4d674974bbb4c5c359db77a6 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/005d0f53e86e4c6fb944362a80a9a0c4 : SUCCESS in 10m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a732e80e5e3044e488552de5f22719fb : SUCCESS in 15m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1ee5f13ab7241a8a706631f7c45bab8 : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a437283060be486ebe76f3d459de8712 : SUCCESS in 11m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/2616a817d3e64c49939e09c0621a09a3 : SUCCESS in 1h 11m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e1fff684015b4aa9b48e995d10853595 : SUCCESS in 1h 17m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aff46f489ffa490791bec16504cf9f54 : SUCCESS in 44m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/deafb8c47cf24e11baccecbd4b5aafb7 : SUCCESS in 7m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0f9b3058343249f387755f0d2322d74b : SUCCESS in 16m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dd522229bad64cba950badf9ead692b1 : SUCCESS in 5m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0775fc21732d44abbbbea527d2750c4e : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fcda90dae4b34a78ac1b850c2ff7d503 : SUCCESS in 6m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/78abf2db6d5c467a9133381ece636a1e : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d592ba1d89fd45a3aa574a494225a45c : SUCCESS in 35m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6c9ced39236d453e86f284507b16060c : FAILURE in 7m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/45594e11d7b84bef8246614a2abcf216 : SUCCESS in 1h 19m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c0ec38f6b9bc4bd1b4ba0f40f03951d5 : FAILURE in 1h 23m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/caab8078f6ca406e95e839d08ddf1e6d : SUCCESS in 1h 32m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/52f6e0f33b654076945a51f89679c26e : SUCCESS in 1h 07m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/037e40714899403d8cc374731ccbfbac : FAILURE in 1h 25m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2cee2b2929334ba894026ef87d92cf22 : SUCCESS in 41m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/34ff3d00fba241029a77355287a823f3 : SUCCESS in 1h 31m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8cc9995bb8d24368b101b39f3f264a30 : SUCCESS in 2h 03m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e02f1f6970a409e8f96ec18bb73a57f : SUCCESS in 57m 47s","accounts_in_message":[],"_revision_number":15},{"id":"a0bb85e142be9e93e8f0456f1dd8b6c7e11b7690","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-05 12:36:52.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/15/upstream-check/cinder-tempest-plugin-ontap-fcp/bed9719/ : RETRY_LIMIT in 11m 28s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/15/upstream-check/cinder-tempest-plugin-ontap-iscsi/368639a/ : SUCCESS in 41m 49s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/15/upstream-check/cinder-tempest-plugin-ontap-nfs/7fd8e6a/ : SUCCESS in 45m 18s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/15/upstream-check/cinder-tempest-plugin-solidfire-iscsi/cbde4c5/ : RETRY_LIMIT in 8m 01s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"e4e5b9dca0b7f1316c7fe5fc3f93126bd0f03d60","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-05 12:50:16.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/82/815582/15/check/DellEMC_PowerFlex/c850ca2/DellEMC_PowerFlex/1143 : FAILURE in 59m 41s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"e26f4227d543a80a33de9043da3e82cf49565be0","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-08-05 13:57:00.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/82/815582/15/check/DellEMC_SC_FC/d7893f4/DellEMC_SC_FC/872 : SUCCESS in 50m 30s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/15/check/DellEMC_SC_iSCSI/9b70cf7/DellEMC_SC_iSCSI/890 : SUCCESS in 2h 06m 28s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":15},{"id":"b9fdc50dffee9ff13b2c04911046cba4f4679de0","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-05 14:12:47.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/82/815582/15/check/DellEMC_PowerStore_FC/db0410b/DellEMC_PowerStore_FC/683 : FAILURE in 15m 07s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/82/815582/15/check/DellEMC_PowerStore_NFS/8b7e53b/DellEMC_PowerStore_NFS/227 : FAILURE in 1h 30m 15s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/82/815582/15/check/DellEMC_PowerStore_NVMeTCP/f71f0d5/DellEMC_PowerStore_NVMeTCP/99 : SUCCESS in 1h 50m 07s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":15},{"id":"71f9d463425ab2e2d3fd9cb0a42fcfa57faa72f3","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-08-05 14:18:01.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/815582-15/HPMSA-iSCSI/ : SUCCESS in 58:37.15\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/815582-15/HPMSA-FC/ : SUCCESS in 58:29.49","accounts_in_message":[],"_revision_number":15},{"id":"5af5f92f77b55f76d06bffa2881c33d94e297511","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-08-05 14:28:28.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/815582-15/Lenovo-iSCSI/ : SUCCESS in 1:05:39\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/815582-15/Lenovo-FC/ : SUCCESS in 1:03:04","accounts_in_message":[],"_revision_number":15},{"id":"97fd3a9bf8ca929c3c00d592cf58fa0c66a3ff02","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-08-05 14:36:08.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/815582-15/PVME-iSCSI/ : SUCCESS in 1:10:35\n- PVME-FC https://os-logs.seagate.com/dell-ci/815582-15/PVME-FC/ : SUCCESS in 1:05:49","accounts_in_message":[],"_revision_number":15},{"id":"04c9f3791ea195571486bffeeac979ecc3fda0c9","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-08-05 14:53:40.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/815582-15/Seagate-iSCSI/ : SUCCESS in 1:14:14\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/815582-15/Seagate-FC/ : SUCCESS in 1:12:31","accounts_in_message":[],"_revision_number":15},{"id":"3407f241e9092614358953301d342f06b1fba776","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-05 16:29:21.000000000","message":"Patch Set 15:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/05_08_2022_14_30_00\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":15},{"id":"69b84c2ea5a045eced777478a813940ace462f81","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-08-06 02:03:13.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/15/20220806083308/NEC-V-FC : SUCCESS in 1h 48m 03s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/15/20220806083308/NEC-V-ISCSI : SUCCESS in 2h 30m 01s","accounts_in_message":[],"_revision_number":15},{"id":"a3c5610d855f653f8b05ea5a6e4c20d5124a1e5d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-06 10:58:39.000000000","message":"Patch Set 15:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/378","accounts_in_message":[],"_revision_number":15},{"id":"183c9d8229fb6e2f1df07651c7582ff1fae28936","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-06 20:36:11.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/15/check/DellEMC_Unity_FC/334c9c8/DellEMC_Unity_FC/1373 : FAILURE in 18m 55s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/15/check/DellEMC_Unity_iSCSI/38a0c3f/DellEMC_Unity_iSCSI/None : RETRY_LIMIT\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"a87d7502d5a97a44fd4cd50b1b943ba391f73790","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-10 21:27:57.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/15/20220811025908/NEC-ISCSI : SUCCESS in 3h 27m 08s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/15/20220811025908/NEC-FC : SUCCESS in 3h 28m 45s","accounts_in_message":[],"_revision_number":15},{"id":"6c7fc9f3359dd6072f58d462c2ca488e5cc9aa33","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:41:03.000000000","message":"Patch Set 15:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":15},{"id":"8a05e1078aacc322af4b48d6e43b1f83aae29659","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:41:34.000000000","message":"Patch Set 12:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":12},{"id":"c83d78d73f344cad580d31adc2b132a1b23ca978","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:41:35.000000000","message":"Patch Set 13:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":13},{"id":"ea43ea2dd07470aa6a97624ef70c4ebe690a7ae1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:42:38.000000000","message":"Patch Set 14:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":14},{"id":"b5cf06c22c68b5642fef83a76be741e428db9f2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-10-17 02:59:45.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"e34aae4080142ca851017d2b53dbd76939982655","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-10-17 02:59:53.000000000","message":"Patch Set 16:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1634 : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"3f149448fbee3b283e102782346b7ab7c0f5c279","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 03:23:05.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2346e533666e44458c693a055d1121fa\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f4e8d460b24d4046b140d501c50b9d41 : SUCCESS in 21m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/90f259e7ab54447593ccf97fcd095fc4 : SUCCESS in 18m 47s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"7654877d07c55edcc92a8ec7be18978da7f67548","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-10-17 04:09:53.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/16/iscsi/ : SUCCESS in 1h 09m 35s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/16/smb/ : SUCCESS in 1h 04m 50s","accounts_in_message":[],"_revision_number":16},{"id":"380d1b669b50ae4600f15b61140928eac8b54efc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-17 04:12:39.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/815582 : FAILURE in 1h 11m 45s","accounts_in_message":[],"_revision_number":16},{"id":"d3913a38e92d096989bc90e77faace868a06a8c8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-17 04:36:51.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/815582 : FAILURE in 1h 35m 42s","accounts_in_message":[],"_revision_number":16},{"id":"ed03754698f8633cfd0c9cf64e69e35d5af029d8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-17 04:37:47.000000000","message":"Patch Set 16:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/76921c8771e44d7e91bbc3b5d4c2d634 : POST_FAILURE in 1h 19m 54s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/f3f73e4147634515a1339a35eb195737 : POST_FAILURE in 1h 34m 51s","accounts_in_message":[],"_revision_number":16},{"id":"ab8fc504aaefaea5db663944380d5b139932e383","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-17 04:48:17.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/16/check/DellEMC_Unity_FC/daa00c5/DellEMC_Unity_FC/1945 : SUCCESS in 1h 09m 08s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/16/check/DellEMC_Unity_iSCSI/7eb2785/DellEMC_Unity_iSCSI/1792 : SUCCESS in 1h 48m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":16},{"id":"330fe348cb6c1785b934edff80d3b471e9fcfe38","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-17 04:52:08.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/16/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/47fca67/ : SUCCESS in 1h 51m 28s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/815582/16/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/eecb458/ : SUCCESS in 1h 34m 07s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/82/815582/16/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/9959036/ : SUCCESS in 1h 31m 35s","accounts_in_message":[],"_revision_number":16},{"id":"86fd31710f91073b3dac3c112f11ba75c440b46b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-10-17 05:18:22.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/82/815582/16/check/dsvm-tempest-cisco-zonemanager-job/fe85050/ : FAILURE in 1h 03m 06s","accounts_in_message":[],"_revision_number":16},{"id":"498dbf1d7d4c8305dedda27f23f6b6f7a877440d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 05:20:01.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e79bb17c7d3409c88b98d2590c92b2b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98b0561696214b88af8428aae6443a12 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29fddfad6b274c939fbfc65ba8c6321e : SUCCESS in 11m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3687ef9ccc1a4128b28176a30939f0ae : SUCCESS in 10m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8a81087c1a5490c9452318685bb6160 : SUCCESS in 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/37bf58d625d54f53b1f40adf2afd841a : SUCCESS in 1h 08m 27s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7d18949f0e2c47408ff96970ecbb551a : SUCCESS in 1h 04m 49s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/83c1a3822b6643b2bcce67f50063de3d : SUCCESS in 1h 39m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61e66ae1fb944ce1951f0900ab73d7e7 : SUCCESS in 50m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2faaf359c3704eadba93edf6e0b1082a : SUCCESS in 9m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b3a1f9a7e29a4d5dbbd3ed78a9d1fca5 : SUCCESS in 16m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/90b6db1c1510469d86506dc5e1baec22 : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/aa8efc7086894e708ecdf2fe1cee03ad : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f33e7d59828144759ff7ecdc13054be3 : SUCCESS in 6m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5bbe712e452141cbbf01a6bd520da0e1 : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6e16df9fd21a45b9a94e088c6974cda1 : SUCCESS in 47m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0bba7fe2c6284674b562b040514833ec : FAILURE in 6m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/777d72feaca84a7bb0b2c213c71f1d86 : SUCCESS in 1h 11m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/353d895de4e643a48a39060a86f3d10c : FAILURE in 1h 15m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2f0e505436fe4c7bae72570b0ddca0c7 : SUCCESS in 1h 42m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/478e1e871130456bbbdf811e637f9a86 : SUCCESS in 1h 04m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4ccff71a73f54ffeae0f8c182b69d1d4 : SUCCESS in 1h 09m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c700be23176b41659e95af7b325805c6 : SUCCESS in 29m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b52a729c681f4121b68da9ad3a3afe8c : SUCCESS in 1h 19m 19s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ba7d971f780f40c38e76a986b38a3d5a : SUCCESS in 2h 13m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89a125550a8e47fa969af1fe428bf4ab : SUCCESS in 37m 56s","accounts_in_message":[],"_revision_number":16},{"id":"0a5a9505e4d22efe6298f9df777b2f00d98ce893","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-10-17 05:27:53.000000000","message":"Patch Set 16:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_10_2022_05_20_11\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":16},{"id":"5a82fc732430715b55dd5ff1953d350625cc78ea","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-17 05:46:22.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/c145fd455215455db78e39892ebf102c : SUCCESS in 1h 33m 04s","accounts_in_message":[],"_revision_number":16},{"id":"e69b602784cc5d1516ddae188ce1e04e68e8cc9e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-10-17 06:00:26.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/815582/16/ : FAILURE in 1h 30m 05s","accounts_in_message":[],"_revision_number":16},{"id":"d9529d01b4916875931747a6e08848f0b1dd4aa8","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-10-17 06:18:30.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/16/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/0ab8b36/ : SUCCESS in 56m 29s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":16},{"id":"fca560402ccb33b3f499dd0675cb8cd8401bb1e6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-17 06:28:07.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/16/upstream-check/cinder-tempest-plugin-ontap-fcp/e72b6a2/ : SUCCESS in 59m 22s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/16/upstream-check/cinder-tempest-plugin-ontap-iscsi/0a939ae/ : SUCCESS in 1h 02m 00s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/16/upstream-check/cinder-tempest-plugin-ontap-nfs/719d766/ : SUCCESS in 42m 17s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":16},{"id":"410881997e4721fcffcd68c23b48f0fd24a63c05","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-17 06:31:21.000000000","message":"Patch Set 16:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-815582-16/ns5_nfs : SUCCESS in 3h 23m 5s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-815582-16/ns5_iscsi : SUCCESS in 3h 31m 4s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":16},{"id":"8c7000863df5c5627e6345991db75ae7fbbe26e0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-10-17 07:02:24.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/82/815582/16/check/DellEMC_PowerFlex/c618d07/DellEMC_PowerFlex/1423 : SUCCESS in 59m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":16},{"id":"95468ce88d6b3666b0edad06015f4dab44aed65f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-10-17 08:06:54.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/82/815582/16/check/DellEMC_SC_FC/214b0e1/DellEMC_SC_FC/1069 : SUCCESS in 47m 43s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/16/check/DellEMC_SC_iSCSI/d55bbad/DellEMC_SC_iSCSI/1053 : SUCCESS in 1h 20m 24s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":16},{"id":"c590c593cf361df795c9bb01f860786806200554","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-17 11:00:57.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/16/check/DellEMC_Unity_FC/e1eb288/DellEMC_Unity_FC/1949 : SUCCESS in 1h 12m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/16/check/DellEMC_Unity_iSCSI/5434098/DellEMC_Unity_iSCSI/1796 : SUCCESS in 1h 55m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":16},{"id":"bd88aa710f4ed0709dfd7b679677725354966461","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-17 12:07:48.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/815582/16/check/iscsi-driver-alletra-6k/162c87f/162c87f.html : FAILURE in 1h 46m 50s","accounts_in_message":[],"_revision_number":16},{"id":"712bc3e5f0908829d7717dae90ed80aab193b6b0","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-10-17 12:52:57.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/16/20221017150945/NEC-ISCSI : SUCCESS in 3h 34m 33s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/16/20221017150945/NEC-FC : SUCCESS in 6h 43m 07s","accounts_in_message":[],"_revision_number":16},{"id":"d82b70ef3363dabe95a471509adcf811e729d103","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-10-17 13:06:37.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/16/check/DellEMC_PowerStore_iSCSI/b159eb7/DellEMC_PowerStore_iSCSI/853 : FAILURE in 1h 09m 36s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/82/815582/16/check/DellEMC_PowerStore_FC/07aa035/DellEMC_PowerStore_FC/1005 : FAILURE in 1h 06m 08s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/82/815582/16/check/DellEMC_PowerStore_NFS/88f8952/DellEMC_PowerStore_NFS/544 : ABORTED in 3h 00m 00s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/82/815582/16/check/DellEMC_PowerStore_NVMeTCP/398ee4b/DellEMC_PowerStore_NVMeTCP/415 : FAILURE in 1h 33m 16s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":16},{"id":"686422fc9b2cf778327b3396ce480e6b3bfedfeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-10-18 10:51:10.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"3da820bc3e55fe76c58e3962ff2a7525cfd2ffc7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-10-18 10:51:22.000000000","message":"Patch Set 17:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1645 : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"b05d8090f4d3c1d5cf5a0e208f0e30de90f5168c","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-10-18 11:24:29.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"e33e08aec1297f390dcb0a2efc732326266f04f9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-18 11:33:16.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/815582/17/check/iscsi-driver-alletra-6k/ee58ee4/ee58ee4.html : FAILURE in 37m 47s","accounts_in_message":[],"_revision_number":17},{"id":"7db20eef848d8cfbf84275e3d740d84f1aad4c05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 11:45:43.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6896f132f1204a2d877a1c07f73598ab\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb4a20dfd22d4ec9a6a080253856efdc : SUCCESS in 28m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eb314729420e4252ab4883fe92efa172 : SUCCESS in 23m 42s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"2d37e5fcf9f2713d71ad154f81bb11c8567b75ae","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-18 11:52:07.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/8538f6634b8c491bac526501f28f747d : FAILURE in 1h 00m 13s","accounts_in_message":[],"_revision_number":17},{"id":"7938c1d734fa8bf1859983a44eb2c13e8310374d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-10-18 11:55:53.000000000","message":"Patch Set 17:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/17/iscsi/ : FAILURE in 1h 03m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/17/smb/ : FAILURE in 45m 31s","accounts_in_message":[],"_revision_number":17},{"id":"2aa5e0010d56790e7f7e85546c9195d04750722e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-10-18 12:14:47.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/82/815582/17/check/dsvm-tempest-cisco-zonemanager-job/bf7824b/ : FAILURE in 1h 03m 26s","accounts_in_message":[],"_revision_number":17},{"id":"18ae3262260323d3e1b4e24821932469fd2bcf4a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-18 12:22:49.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/815582 : SUCCESS in 1h 30m 25s","accounts_in_message":[],"_revision_number":17},{"id":"684d4a99ed580103720d82e872a8b4ad3a21c284","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-18 12:22:53.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/815582 : SUCCESS in 1h 29m 25s","accounts_in_message":[],"_revision_number":17},{"id":"cd73fdae37458a3709280e3faabaa0271b4bb198","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-18 12:33:34.000000000","message":"Patch Set 17:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/5a8b268b689d4c518c802e8a8250df3b : POST_FAILURE in 1h 28m 08s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/191bf07058c44bb39cfaf0411c1b0fd0 : POST_FAILURE in 1h 36m 01s","accounts_in_message":[],"_revision_number":17},{"id":"15f34b84a980e4c23ce81c5ce3019426832f5daf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-18 12:44:00.000000000","message":"Patch Set 17:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/17/check/DellEMC_Unity_FC/454b9a9/DellEMC_Unity_FC/1955 : SUCCESS in 1h 18m 30s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/17/check/DellEMC_Unity_iSCSI/4ea4139/DellEMC_Unity_iSCSI/1802 : SUCCESS in 1h 52m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":17},{"id":"d04259e4dcc8684b4ca729102e0b6c5dbfafedf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 12:49:59.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f17724102d45471194b8c1b8695dbbef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97fa1574731f462d86b4ce70820d8e3e : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a3c896cbe7b462da1702627139057b1 : SUCCESS in 10m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2035e3e8e7a4049b8e5650c9cb678b0 : SUCCESS in 10m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfc24a66e89a482987fa2a4417e144c8 : SUCCESS in 10m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/641433e742d943c3823894c761dfe4e8 : SUCCESS in 50m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/939c06a09f214fb289f02a2f55ccf399 : SUCCESS in 46m 17s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f3810aeb65b944c0a915efa765e75ff4 : SUCCESS in 1h 14m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97f529dc19ab4701ba1234606ffd10a6 : SUCCESS in 38m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6e2466e01344a129c02c69f8e6e08f5 : SUCCESS in 9m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/efe46faa9f9b4a929ed80b352cac2541 : SUCCESS in 14m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d25b37b2de2745fda306161e41983c87 : SUCCESS in 5m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f4e9fc480f3c439fa21fd52aed89dc24 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8537980e01b6451eab425bccc35bbcb3 : SUCCESS in 5m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/67b43cac12bf4927ba3b6376a04750e4 : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cc75ab95a8c942e2968d7251250f59de : SUCCESS in 37m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0f68d863214e4e1e9a43f228389430c4 : FAILURE in 6m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d131c5705808413d869bdc17a1c51891 : SUCCESS in 1h 08m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bb8482733fff410997ce99af684ca6f2 : FAILURE in 1h 23m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f19fb30a762249979c0a2bf7f5baef7c : SUCCESS in 1h 32m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d0a4bafa041348eaabcee9ac3c77832e : SUCCESS in 1h 02m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a75e3796f59a450a85d97ef197c1748f : SUCCESS in 1h 20m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6082e619c1fe4e53b1460024ea0ba94c : SUCCESS in 31m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/418ae4f17c4944ebb5f36d13d2b695b7 : SUCCESS in 1h 23m 43s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a63dd8da856b4887836959894045372f : SUCCESS in 1h 56m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2dbdfd975da34b2b840ce4de7dc97c84 : SUCCESS in 43m 42s","accounts_in_message":[],"_revision_number":17},{"id":"225df39cc4a3b357ec463d3d3d2ac737c14752ec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-10-18 13:05:16.000000000","message":"Patch Set 17:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_10_2022_12_57_42\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":17},{"id":"9f15e6de9a338b534b3f4e516e7462db59bfd712","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-18 13:14:19.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/17/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8e84a80/ : SUCCESS in 1h 32m 45s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/815582/17/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/29b416f/ : FAILURE in 1h 47m 16s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/82/815582/17/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/718687d/ : FAILURE in 2h 21m 44s","accounts_in_message":[],"_revision_number":17},{"id":"7a9b8a30a5a76e9400dbcf326775257b3d64b5b2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-10-18 13:46:20.000000000","message":"Patch Set 17:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/82/815582/17/check/DellEMC_PowerFlex/44d13cc/DellEMC_PowerFlex/1426 : SUCCESS in 55m 55s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":17},{"id":"255286ceb0a1181efbc79b686e55cd751650d68f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-18 14:11:32.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/17/upstream-check/cinder-tempest-plugin-ontap-fcp/afa4887/ : SUCCESS in 1h 00m 35s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/17/upstream-check/cinder-tempest-plugin-ontap-iscsi/d70652c/ : SUCCESS in 1h 15m 13s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/17/upstream-check/cinder-tempest-plugin-ontap-nfs/e8c9739/ : SUCCESS in 1h 11m 52s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":17},{"id":"c5b2ae5cb0e9a49b9791ee7b290bd636dcaf06c2","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-10-18 14:13:29.000000000","message":"Patch Set 17:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/82/815582/17/check/DellEMC_SC_FC/eba3ca3/DellEMC_SC_FC/1071 : SUCCESS in 50m 57s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/17/check/DellEMC_SC_iSCSI/5d97e9e/DellEMC_SC_iSCSI/1055 : SUCCESS in 1h 23m 12s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":17},{"id":"e5ce6b48417a8511d211c6823148b1f8ed97d689","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-10-18 14:21:33.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/17/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/d0929d7/ : SUCCESS in 1h 25m 13s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":17},{"id":"7840349fac4cf3e3541643bc2527549f6326eacd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-18 14:21:46.000000000","message":"Patch Set 17:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-815582-17/ns5_nfs : SUCCESS in 3h 27m 31s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-815582-17/ns5_iscsi : SUCCESS in 3h 29m 58s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":17},{"id":"f26dc16ea724ce0b1c071c8281c194d4c5a8b49d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-18 14:39:02.000000000","message":"Patch Set 17:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/17/check/DellEMC_Unity_FC/1e1513d/DellEMC_Unity_FC/1956 : SUCCESS in 1h 10m 45s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/17/check/DellEMC_Unity_iSCSI/8fc7f00/DellEMC_Unity_iSCSI/1803 : SUCCESS in 1h 48m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":17},{"id":"b5e15cf2bd6c7fe6c44f51b636816dd9f0243511","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2022-10-18 16:00:06.000000000","message":"Patch Set 17:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/82/815582/17/fc : SUCCESS in 1h 22m 40s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/82/815582/17/iscsi : SUCCESS in 1h 37m 16s","accounts_in_message":[],"_revision_number":17},{"id":"36b053a01e1b7f01911747524f74e7da2ce3f584","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-10-18 23:40:26.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/17/20221019033003/NEC-FC : SUCCESS in 3h 22m 58s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/17/20221019033003/NEC-ISCSI : SUCCESS in 5h 10m 20s","accounts_in_message":[],"_revision_number":17},{"id":"ea57851553e31fc3cf100012ecad422b0e71299e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-10-20 12:21:59.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/815582/17/ : SUCCESS in 1h 25m 52s","accounts_in_message":[],"_revision_number":17},{"id":"ac909f4636925788db92129c21b6de69bd47aa3f","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-10-20 14:06:53.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"c31a5fe009878222e901bb431aa7ad47e2db19fd","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-10-20 15:00:03.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"3efb7eebb04ed7968ca68bae1b18d9280dec0c55","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-10-20 15:05:10.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"0eaa930016143df276a3a73d531cf8de6665627b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-10-20 23:20:04.000000000","message":"Patch Set 17:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/766","accounts_in_message":[],"_revision_number":17},{"id":"5cc0a9e1a7998e62d514ad71d3cd1295a457e77c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-10-25 01:49:30.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/17/20221025040947/NEC-V-ISCSI : SUCCESS in 2h 21m 57s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/17/20221025040947/NEC-V-FC : SUCCESS in 6h 39m 41s","accounts_in_message":[],"_revision_number":17},{"id":"f4444f0d25803daa06705371d02bf10ab198c8a4","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-10-26 10:06:12.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"25b746e5d4d78e844103eb92e473bed9bac39a8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-10-26 10:26:02.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"634a912ff4d14f04406e17d80d40ff0a51a6dbaa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 10:50:42.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34d91b72ea944f248e047bd9cb87a111\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2dda18c161f4e3c97e599568e2e698e : SUCCESS in 22m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/28e9c2c02f0240ffad4dd47781ee08c3 : SUCCESS in 20m 34s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"3e3dff29abd6d58dda43f9d73eed435d4f1111fd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-26 11:35:39.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/815582/18/check/iscsi-driver-nimble/ad90caa/ad90caa.html : FAILURE in 12m 30s","accounts_in_message":[],"_revision_number":18},{"id":"b463213dbe79e221e6b2faeadefc6dd8d271f9d3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-10-26 11:38:15.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/18/iscsi/ : SUCCESS in 1h 11m 32s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/18/smb/ : SUCCESS in 49m 48s","accounts_in_message":[],"_revision_number":18},{"id":"f5fa87b42e411cc602a9a1c90c01e5b88655cc0b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-26 11:39:58.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/815582/18/check/iscsi-driver-alletra-6k/5a18b98/5a18b98.html : FAILURE in 1h 09m 01s","accounts_in_message":[],"_revision_number":18},{"id":"332f26ce81b3d1679b37a107e588ba227fd456f2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-26 12:09:37.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/18/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/202fe63/ : SUCCESS in 1h 38m 07s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/815582/18/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6d8bfef/ : SUCCESS in 1h 37m 21s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/82/815582/18/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/76572af/ : SUCCESS in 1h 42m 27s","accounts_in_message":[],"_revision_number":18},{"id":"d145ef13010591c3d08f7bc94e2aa84cf45b9475","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-26 12:23:36.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/7825f33c0e9845809f3f3c9a7bfacb40 : SUCCESS in 1h 04m 36s","accounts_in_message":[],"_revision_number":18},{"id":"8f37accfed539062f05ede65933747889178a11e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-10-26 12:26:12.000000000","message":"Patch Set 18:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1666779973032 : SUCCESS","accounts_in_message":[],"_revision_number":18},{"id":"759fec49c4a25fab687a6fa7a69f0732fa2ac04a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-26 12:29:25.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/815582 : SUCCESS in 2h 01m 36s","accounts_in_message":[],"_revision_number":18},{"id":"b2ab049dd78082bee71a4ce317bdcf26a64563cf","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-26 12:30:00.000000000","message":"Patch Set 18:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/d5d9de7b292f4ca596009d6d2733d6f3 : POST_FAILURE in 1h 39m 21s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/b2e29b4a29a742eca65648da178992db : POST_FAILURE in 1h 54m 28s","accounts_in_message":[],"_revision_number":18},{"id":"1c5fc2590fbe5874034367aeeca57fc37a5e7a4a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-26 12:38:53.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/815582 : SUCCESS in 2h 11m 20s","accounts_in_message":[],"_revision_number":18},{"id":"8001f71cf242a17b540e412ceb527dc8505e6254","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 12:45:54.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2e0b6bdb85f4f1da29c57c282347ff0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ebfe9866e7149f893b995f4381a2bd3 : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e93dcb933cd4ed3a7aecffe10411aec : SUCCESS in 10m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd1f8c85617e44c29da244f0d5629216 : SUCCESS in 11m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ffe33725e904df5afaea3dd00c87b8d : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/cd6f0a1b439a4f5c819479b4cd924c3e : SUCCESS in 58m 20s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5add2677b2fc403ba0cb0a41c18620e8 : SUCCESS in 1h 00m 39s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/045787c133f54a309f589f8cc3a9dea4 : SUCCESS in 58m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3b72732f90b4bcb9fc7fe945907845f : SUCCESS in 53m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7e45ddd1f6d42a3a36e29f90f48140f : SUCCESS in 7m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/daf47761f998494a8b6c29d9bd11437e : SUCCESS in 12m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/09118a29b7964ab7a91e382c66cad022 : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/326668f2293e45fa93781cd17ff18f76 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b0873993ea1e486eb0a5b6cf988e883a : SUCCESS in 5m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ab30792ada0042b682266fc13a0f7331 : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ffcc22f90e224f68b20dfbdae420b6e7 : SUCCESS in 29m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/61e5e10aeacd4be0b51b430e5aa67712 : FAILURE in 5m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c563d4f3092c4b93a5e2793370ca7720 : SUCCESS in 1h 16m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/87a5380159fb41dd8e80ab3ee1699943 : FAILURE in 1h 30m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8653c4b1f8494147856881b028dce378 : SUCCESS in 1h 38m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dae63fb81d17475980ae14c22eeed63b : SUCCESS in 1h 00m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/094d1bb31f8c48788dca5662b85f53a4 : SUCCESS in 1h 32m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/65c4755b7987435a960aba840be4c350 : SUCCESS in 46m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5deb4cd381d24cf3aff0da414a7cd71b : SUCCESS in 1h 15m 32s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fccc37de15eb480288b1f302f702fc86 : SUCCESS in 2h 13m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6450c3e1a8f4a448649b89bb65ddecf : SUCCESS in 38m 21s","accounts_in_message":[],"_revision_number":18},{"id":"bb1e95fa13e336ad7363bb410f1cd7b0dd93e787","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-10-26 12:54:08.000000000","message":"Patch Set 18:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/26_10_2022_12_46_02\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":18},{"id":"6c85eeb994f7da5983e0b9d4d62d216cd01df939","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-10-26 13:35:05.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/815582/18/ : SUCCESS in 1h 34m 03s","accounts_in_message":[],"_revision_number":18},{"id":"f897bdb90b6fd9dfa3c3e850f7dfcaf1d6d83d30","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-10-26 13:41:01.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/18/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/0d17aee/ : SUCCESS in 53m 35s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":18},{"id":"4926d05958d8fc25aa525064b8fb08d28f30ff44","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-26 13:57:41.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/18/upstream-check/cinder-tempest-plugin-ontap-fcp/3fb6869/ : SUCCESS in 56m 13s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/18/upstream-check/cinder-tempest-plugin-ontap-iscsi/9f7e0c4/ : SUCCESS in 1h 09m 25s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/18/upstream-check/cinder-tempest-plugin-ontap-nfs/a859711/ : SUCCESS in 1h 10m 14s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":18},{"id":"98f5866d1fa1e2a56d9c59e04676fc2c4d2514ae","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-26 14:08:51.000000000","message":"Patch Set 18:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-815582-18/ns5_nfs : SUCCESS in 3h 29m 30s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-815582-18/ns5_iscsi : SUCCESS in 3h 30m 45s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":18},{"id":"23181a2a4a69ada474803dae5f33a81fc8a23f8a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-10-26 14:26:35.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/82/815582/18/check/DellEMC_PowerFlex/52bd2c4/DellEMC_PowerFlex/1438 : FAILURE in 1h 00m 38s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":18},{"id":"3d52e2fb2a863943fc55ad6ad3cb4e8e2525d377","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-10-26 15:03:09.000000000","message":"Patch Set 18:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/82/815582/18/check/DellEMC_SC_FC/d84ef37/DellEMC_SC_FC/1083 : SUCCESS in 56m 48s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/18/check/DellEMC_SC_iSCSI/f420fff/DellEMC_SC_iSCSI/1067 : SUCCESS in 1h 15m 20s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":18},{"id":"0d7fbdfa74ddccb4e5cd0855b37668cafd12c833","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-10-26 15:56:48.000000000","message":"Patch Set 18:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/18/check/DellEMC_PowerStore_iSCSI/2780655/DellEMC_PowerStore_iSCSI/865 : SUCCESS in 1h 21m 46s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/82/815582/18/check/DellEMC_PowerStore_FC/99b1ae7/DellEMC_PowerStore_FC/1017 : SUCCESS in 1h 04m 05s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/82/815582/18/check/DellEMC_PowerStore_NFS/f60dd2b/DellEMC_PowerStore_NFS/556 : SUCCESS in 41m 20s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/82/815582/18/check/DellEMC_PowerStore_NVMeTCP/0444f18/DellEMC_PowerStore_NVMeTCP/427 : SUCCESS in 1h 42m 00s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":18},{"id":"058681b6bdae578d6d7c3d5b889801c92bc90f77","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-26 16:15:34.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/18/check/DellEMC_Unity_FC/174eca4/DellEMC_Unity_FC/1979 : SUCCESS in 1h 12m 20s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/18/check/DellEMC_Unity_iSCSI/c56c9b4/DellEMC_Unity_iSCSI/1826 : ABORTED in 3h 00m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":18},{"id":"bb2cd1f784f55a5f490a9ee132888e8444d6cc95","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-10-26 17:26:30.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/18/20221026225159/NEC-FC : SUCCESS in 3h 32m 17s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/18/20221026225159/NEC-ISCSI : SUCCESS in 3h 34m 28s","accounts_in_message":[],"_revision_number":18},{"id":"eea4de7c9be0f618e1887f1aadf4ff0a795a0f67","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2022-10-26 19:19:56.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"b9b6352089ab74ca360ce08dee61286f0c4b975a","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2022-10-26 20:38:47.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"19af7b7011dae1e13478526c85e105c04e8c6557","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-10-27 00:42:29.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"02e9f8415dcf0abd1ee8b14fe1fbed46c28e40f1","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-10-27 00:43:40.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"4c3464794a89404771d753b66b915fd4320bfe92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 00:48:55.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/436644780d924ff2936eb59fb0a6dc60\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f65c4ef1bb844cb3b8b9109cde94567e : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4ab227f55ddf4737bdfb5f34fe9ad389 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"d59d96dd496937dd44b3ed7b4b6ae92eb9ba4aef","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-27 00:58:09.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/82/815582/19/check/iscsi-driver-nimble/529416b/529416b.html : FAILURE in 12m 41s","accounts_in_message":[],"_revision_number":19},{"id":"9aa402a399129aa7126507137731e620477f66ef","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-10-27 01:12:04.000000000","message":"Patch Set 19:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1666831359526 : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"5cd95da9d6ab481f79db7d33f2d4b821c63bcc91","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-10-27 01:49:09.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815582/19/iscsi/ : SUCCESS in 1h 05m 46s\n- cinder-smb http://cloudbase-ci.com/cinder/815582/19/smb/ : SUCCESS in 55m 10s","accounts_in_message":[],"_revision_number":19},{"id":"5267e87800bcd6208d8000242e35769e20c6f9fe","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-27 01:52:00.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/82/815582/19/check/iscsi-driver-alletra-6k/09180d8/09180d8.html : FAILURE in 1h 07m 03s","accounts_in_message":[],"_revision_number":19},{"id":"ce7f2d937aa906f5a70df5bd1a2b17b3bdb884b6","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-10-27 02:15:54.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/82/815582/19/ : SUCCESS in 1h 22m 30s","accounts_in_message":[],"_revision_number":19},{"id":"0fec7b71eaff03f5e3fb6f0c675f1ba3b788011c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-27 02:22:11.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/815582 : FAILURE in 1h 36m 15s","accounts_in_message":[],"_revision_number":19},{"id":"9bcf417c5d9bf34b42024515b76c627c68c7eaac","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-27 02:28:05.000000000","message":"Patch Set 19:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/837fd30fbce949f0a54537c671f6c415 : POST_FAILURE in 1h 29m 05s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/6bb789cbf52b4ed9aca5d01cbd283bb0 : POST_FAILURE in 1h 38m 39s","accounts_in_message":[],"_revision_number":19},{"id":"76dd685efdb74ac3b13ab8e04baac0d549445530","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-27 02:34:39.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/815582 : SUCCESS in 1h 48m 19s","accounts_in_message":[],"_revision_number":19},{"id":"3082709f807e7ebee2f8b5e2bf85c60c1fbdca4f","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-27 02:37:15.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/3c8d6849cbc04adc85d5b3eb5d52e02f : SUCCESS in 1h 51m 26s","accounts_in_message":[],"_revision_number":19},{"id":"c3a5270f9c18cb5802d1d0b289b98406f6bac5f4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-27 02:40:07.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/82/815582/19/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a88e23a/ : SUCCESS in 1h 34m 13s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/82/815582/19/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/c007894/ : SUCCESS in 1h 55m 49s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/82/815582/19/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/f0556d8/ : SUCCESS in 1h 56m 08s","accounts_in_message":[],"_revision_number":19},{"id":"b702ffb28c9e37ffe6ce07a9358898674b310773","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 02:53:24.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/feebf157a96e415ea8cfe13d43c925e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c4ddc1d8a42465d8da1774aadf9fa43 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ffb5fefca6f410ca21db668ae1ec42a : SUCCESS in 9m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b90c469f4fc54e8daa3285b4c5deb208 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f90905c281b451589b927960d057c85 : SUCCESS in 14m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/0c4689164e8c411abdf2dd2abd6384b3 : SUCCESS in 42m 44s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8375e3932a0b4d92a5b7b937c383ccec : SUCCESS in 43m 49s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/63100a709d164a27929b4ad0a02d0e15 : SUCCESS in 58m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2177ec8ecba8459982afa7208463c10a : SUCCESS in 35m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b5c45f1a8c74256b5e0784576d223ae : SUCCESS in 7m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1bf287d072d843fba3cfce5ad7c9709a : SUCCESS in 12m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6ce0db46529442a5a934c115c3bc2f31 : SUCCESS in 5m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4513efa01bc04093b0f1bdedbbcb86d5 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cf2933ba45304ff8868e1a07711dbd48 : SUCCESS in 6m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/96bb13e86d664c72b7888baa727477cc : SUCCESS in 7m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5863ceb2697e4ac783260113f54a353b : SUCCESS in 27m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f4b241ca01b7419988b92c2c89a757e5 : FAILURE in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f319f8f16001404eb380cdc16eb71ded : SUCCESS in 58m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5bd6192197fe4595a911ca8679632b8d : FAILURE in 1h 26m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/effd1380b7cb44f49e0f2c276b37c7bf : SUCCESS in 1h 10m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/83325e95b72245d2a2c642f32e0491b4 : SUCCESS in 1h 00m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6facafb34c41476c9d737e8c44a533ad : SUCCESS in 1h 30m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/48a57e2a954a426092d7cb3cdfdf6276 : SUCCESS in 43m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/15342cc34c8d476394ec82bc89037b69 : SUCCESS in 1h 32m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f4ec93a44da94e91927a5586cfaf2c1a : SUCCESS in 2h 02m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/765246638169497795067c792f958023 : SUCCESS in 45m 55s","accounts_in_message":[],"_revision_number":19},{"id":"1da3f5f946fcfb58e299a0d044d1bc36bba43be6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-10-27 03:01:38.000000000","message":"Patch Set 19:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/27_10_2022_02_53_34\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":19},{"id":"e0c95a1d1782d3343802f62657fb014f3cbd59fb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-27 03:42:59.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/82/815582/19/check/DellEMC_Unity_FC/99f7b5b/DellEMC_Unity_FC/1980 : SUCCESS in 1h 15m 27s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/19/check/DellEMC_Unity_iSCSI/bfb589a/DellEMC_Unity_iSCSI/1827 : ABORTED in 3h 00m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":19},{"id":"ffc82296554036c54bc1cd8e104978816c28364e","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-10-27 03:47:54.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/19/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/179cb7d/ : SUCCESS in 52m 36s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":19},{"id":"8597becdbacc3c9746e0821047706e7289e49411","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-10-27 03:54:28.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/82/815582/19/check/DellEMC_PowerFlex/d44ee31/DellEMC_PowerFlex/1439 : FAILURE in 1h 00m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":19},{"id":"13559e344fd1f8a98b49816513d9747e549115c8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-27 04:01:40.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/19/upstream-check/cinder-tempest-plugin-ontap-fcp/d6d556e/ : SUCCESS in 1h 02m 47s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/19/upstream-check/cinder-tempest-plugin-ontap-iscsi/33a5a67/ : SUCCESS in 1h 01m 02s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/19/upstream-check/cinder-tempest-plugin-ontap-nfs/abcf76e/ : SUCCESS in 43m 57s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":19},{"id":"4a120115ca5b5bb7ed13db31185b2eeb1026dd70","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-10-27 04:12:07.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/82/815582/19/check/DellEMC_SC_FC/55eff59/DellEMC_SC_FC/1084 : SUCCESS in 49m 15s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/19/check/DellEMC_SC_iSCSI/b01562a/DellEMC_SC_iSCSI/1068 : SUCCESS in 1h 18m 22s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":19},{"id":"d18ce80753c814b5a150aab5816180453e8342d5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-27 04:16:33.000000000","message":"Patch Set 19:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-82-815582-19/ns5_nfs : SUCCESS in 3h 24m 24s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-82-815582-19/ns5_iscsi : FAILURE in 3h 33m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":19},{"id":"5bb1098a22cab20d553102fb94b2aed3fc1476a9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-10-27 04:17:00.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/19/20221027094231/NEC-FC : SUCCESS in 3h 33m 52s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/19/20221027094231/NEC-ISCSI : SUCCESS in 3h 34m 26s","accounts_in_message":[],"_revision_number":19},{"id":"10ba9b26b140cf6ac5d362932eeb72d38b221437","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-10-27 04:29:03.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/82/815582/19/check/DellEMC_PowerStore_iSCSI/0996d02/DellEMC_PowerStore_iSCSI/866 : SUCCESS in 1h 24m 03s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/82/815582/19/check/DellEMC_PowerStore_FC/09d18b6/DellEMC_PowerStore_FC/1018 : SUCCESS in 1h 02m 36s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/82/815582/19/check/DellEMC_PowerStore_NFS/c37014a/DellEMC_PowerStore_NFS/557 : SUCCESS in 40m 33s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/82/815582/19/check/DellEMC_PowerStore_NVMeTCP/863600b/DellEMC_PowerStore_NVMeTCP/428 : SUCCESS in 1h 35m 15s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":19},{"id":"895758caa7e478385f8b2becae59c3cc4cb761e5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-10-28 05:52:23.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/19/20221028090109/NEC-V-ISCSI : SUCCESS in 2h 40m 05s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/82/815582/19/20221028090109/NEC-V-FC : SUCCESS in 5h 51m 10s","accounts_in_message":[],"_revision_number":19},{"id":"bd264ed03560bc91bcc48b0645d36cffea7a39f8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-10-28 09:32:14.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 11h 28m 47s","accounts_in_message":[],"_revision_number":19},{"id":"3f46cffc88ee528100a70e6bdb796a0ac0ae85fc","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-10-31 08:35:51.000000000","message":"Patch Set 19: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"da4c4b972a0ec502fba062dd97fde24be3202fb8","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2022-11-02 00:07:50.000000000","message":"Patch Set 19:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/fc : SUCCESS in 1h 07m 51s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/iscsi : FAILURE in 1h 18m 51s","accounts_in_message":[],"_revision_number":19},{"id":"3279a4e8821d4c753a6e7a8d9ab968698a2f476f","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2022-11-02 00:19:36.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"b55e7ee57b47ff3abf58264fbdd8eae3fcef7f9a","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-11-30 14:26:01.000000000","message":"Patch Set 19: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"9db0662f01d8b64ee1d68200a6d6c4598c7d6db8","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2022-12-01 19:03:21.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"f91e99400728193b2c7e0b107e915245ec21e6d7","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2022-12-16 02:38:16.000000000","message":"Patch Set 19:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/88/864288/1/fc : FAILURE in h m 0s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/88/864288/1/iscsi : FAILURE in h m 0s","accounts_in_message":[],"_revision_number":19},{"id":"72b40b2bbd57c807dcf13fdcd1122f925872116c","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2022-12-16 03:35:03.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"06d614591042b3483885a7b67d61427234948c3c","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2022-12-20 11:19:39.000000000","message":"Patch Set 19:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/88/864288/1/fc : FAILURE in h m 0s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/88/864288/1/iscsi : FAILURE in h m 0s","accounts_in_message":[],"_revision_number":19},{"id":"0a8753fd228273c9d650dd9bdcb0230402f473df","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2022-12-22 07:20:00.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"ad4102fca15023fe42f7e417dd3f753b151600fe","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-12-26 23:33:17.000000000","message":"Patch Set 19: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"2595eb3224b5eb4a6882d1accb48998db0ab1c74","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2022-12-27 04:05:12.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"b7319a5b7d28dc8269667a6af862f82682925878","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-01-03 15:20:43.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"bba4feb5d5d8f1f20bd40bf01a83f9873ee11422","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-01-04 14:14:08.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"581babb081b0d47023af721eef84f0249bead345","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-05 00:15:35.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"2f0d7586968e39c35db49b4ec4812c4a209a59ba","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-01-11 08:11:08.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"e7a75e38ef8b3959753b89ec22bec122d73c5b01","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-11 10:35:37.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"318a64ba5c4b62768abd5467cb9cda791300ff82","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-12 05:33:47.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"ba65704b8884e0507ce147e7c34570e4a2e74975","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-12 05:34:15.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"5765c1c827c6b8d3929289ada3d8dbcb8e84cf8e","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-01-12 11:00:07.000000000","message":"Patch Set 19:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/fc : SUCCESS in 1h 07m 51s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/iscsi : FAILURE in 1h 18m 51s","accounts_in_message":[],"_revision_number":19},{"id":"9614e517914481b32d27aa875416ecddc0a6a58d","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-12 17:57:52.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"0333d98c7725302d9c1519002a96d42d0ad318ed","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-01-13 04:00:05.000000000","message":"Patch Set 19:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/fc : SUCCESS in 1h 07m 51s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/iscsi : FAILURE in 1h 18m 51s","accounts_in_message":[],"_revision_number":19},{"id":"c90da6ca6ab0ad457557f4692f12dcedf1564ae7","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-13 04:58:23.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"1d9f8a956be29b89cd4056d5b2753192b22a1d97","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-13 05:06:14.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"598762c19367a6e8e186dc74fb875dc6946e0e53","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-01-13 10:38:20.000000000","message":"Patch Set 19:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/zyxjek/fc : FAILURE in 1h 31m 24s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/zyxjek/iscsi : FAILURE in 4h 51m 13s","accounts_in_message":[],"_revision_number":19},{"id":"c3b6ce80ebf0890b7df5483e979a84b916907976","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-01-17 17:03:01.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"0058fbb6b2fa04c577e6298dc0d6d56c4a5c8361","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-18 23:05:14.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"504adf860ed7c66c67f8cf7befd5e25171b3af9d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-01-20 05:41:35.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"2844112bfa44418c9f824b3884218c1339ede8f8","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-20 05:46:30.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"695f96ad30d89bfe7547407f40a24b8d5d38d543","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-01-20 15:02:23.000000000","message":"Patch Set 19: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"6072b052f18ca0541fe821dd0c7418878f8ffa78","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-01-20 16:51:07.000000000","message":"Patch Set 19: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"6c7b7551f33e9fcff82ed2c2107ea055e33630ed","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-21 11:16:56.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"a63abca2af7734114bf434e6ac73fc02c8ed251d","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-21 11:20:36.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"414cc779e4a682352250e2386939309c41439d26","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-21 11:36:55.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"18b7ab4c9e4df02e2f68cf090bff804caddfbf47","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-21 11:37:35.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"8c3274dbe604c725eeb3359af217fed6d4243924","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-01-21 13:44:09.000000000","message":"Patch Set 19:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/alfpnn/fc : FAILURE in 1h 31m 46s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/alfpnn/iscsi : SUCCESS in 1h 49m 32s","accounts_in_message":[],"_revision_number":19},{"id":"908dc99efcc4e4d1987f2c58974b869a82f5a212","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-21 14:01:14.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"181e45a4afcf3c116a4fbb6d1c6d16c25beefe40","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-22 07:25:25.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"16ec17d8f62dc5719f7485a90ae48fa19bdb79a5","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-01-22 11:31:45.000000000","message":"Patch Set 19:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/tazihu/fc : SUCCESS in 3h 16m 40s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/tazihu/iscsi : SUCCESS in 3h 32m 46s","accounts_in_message":[],"_revision_number":19},{"id":"8f5106b8f8c9997eb2327bbcc4b7318381a79e63","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-22 11:33:40.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"a81670204e16cdb45b3b51e832a78caa17770116","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-23 16:56:10.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"bd4f39eca6306c504b55beaf18a546032a2b6824","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-01-24 17:25:37.000000000","message":"Patch Set 19: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"06f0bd623a4b2dc4012472ce12adf81322fb6d28","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-24 22:16:26.000000000","message":"Patch Set 19: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"8d1331e71f20a55faffd6be1c914d9de888b8ec8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-24 22:35:01.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"885f812b968fd3b7dd9e05eb402707fcbeb13c69","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-24 22:38:18.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"9d07a73d0ca142730244a67da67e5108cb879cab","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-01-25 10:00:40.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"f783ce62c93ad419b5f265c75e65a301479f78df","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-26 01:43:10.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"587c56c9766b4cf1e122b211b538db835729b197","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-01-27 02:05:04.000000000","message":"Patch Set 19:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/jwokdb/fc : SUCCESS in 3h 16m 40s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/jwokdb/iscsi : SUCCESS in 3h 32m 46s","accounts_in_message":[],"_revision_number":19},{"id":"c13d7496696fd44fbffbf3b440d3355e23925312","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 03:01:53.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"ac527777ec16b4db1aa1e9de227338e5aac53b24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 04:28:54.000000000","message":"Patch Set 19: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/846977 is needed.","accounts_in_message":[],"_revision_number":19},{"id":"3f6586fd1a6d8e9d6c44f83a8beba0d0d8c95fae","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-01-27 08:42:34.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"6c59f8da2c1e7e0a5866a51225ac103db9aabe08","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-27 08:44:18.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":19},{"id":"c09a9bacdbba863e7873774da3cb7c4d76649683","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-01-27 08:53:54.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/ab9f8e73d6d741e095fec14f40614404 : POST_FAILURE in 4m 29s","accounts_in_message":[],"_revision_number":19},{"id":"b84d80ee8a39d31e8767448939a59da4389e3d05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 09:12:47.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d5a915a35c648298a5e4efcb5980478\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c74a9b0eaaf84ba0b7ebfaad7700aea8 : SUCCESS in 14m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1518b083bc6248f5b5f9455a26e01c11 : FAILURE in 27m 53s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"1a2f4e1024b582c571e44d1eb2a2fe12fcfe913c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-27 10:01:34.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/51ad04d9816c4b64a39cdce5a2806c60 : SUCCESS in 1h 17m 20s","accounts_in_message":[],"_revision_number":19},{"id":"65dbbad8d9c1a81fdf14a63c8ba645f98c5a0913","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 11:01:39.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d8febe918a745f381afae14a53aeda1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56f28ea76e6148b7a86a0e9c0e6add45 : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18468a1fc0184e39abdc6fa1779e14f1 : SUCCESS in 12m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f3bd32868cf43fdbf50133b678f6f1d : SUCCESS in 11m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30af7435cf2e4740a3a5860aa19ce777 : SUCCESS in 12m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/28a8c2e7bd184614b7caa7c7ff9ceffd : SUCCESS in 55m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0532aa9b046d4e4eac1f59e8c670a1b4 : SUCCESS in 1h 09m 42s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bdda5eb7b5ea4cc5b0bde7bca1d0fd42 : SUCCESS in 1h 44m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99add77cc32147c596418f3d66872769 : SUCCESS in 1h 00m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8108735c14e42a190de6b6185cacee0 : SUCCESS in 7m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5e4a314ceeb84c50a5e0b38694e1299b : SUCCESS in 16m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cbf4310e6e22433eb85039d23c547f64 : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/158e35455b8c4da1a48c50a645cdd0e5 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/16a579fecd1d4757bd9bfe4869e80b5f : SUCCESS in 5m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8da6209f30744a7087ec71daba23d605 : SUCCESS in 5m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9396d670b9054a388be532c5c208139d : SUCCESS in 46m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/01bd7529f8b54f7eb34c772953aef782 : FAILURE in 4m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/852785ced3924240af9baaac8d91b75c : SUCCESS in 1h 10m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9a4b847aa05143f08ce8f7814face547 : FAILURE in 1h 17m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9d9d7d32f2484051909e3194779c1a8b : SUCCESS in 1h 42m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/890e4fc264eb459d84b971eee753e769 : SUCCESS in 1h 20m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f51e316cde414daf873216a50bc483bf : SUCCESS in 1h 23m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b2f4b22c7b2b4adea8dc8afb3cf2dedd : SUCCESS in 43m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1827be19983b404baf402f611f8aa2e6 : SUCCESS in 1h 18m 46s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/03733697a4c74db78649b499b438ff8e : SUCCESS in 2h 03m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24799995d92d4044a2ace9c3db626ef2 : SUCCESS in 51m 42s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bafc4b2e7b60488e8d5a6d60b60df36c : SUCCESS in 1h 22m 07s","accounts_in_message":[],"_revision_number":19},{"id":"6c1955883def11734260125c639e6b1dc330faa7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 11:02:31.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"2a718ce15f47424a07041f2abf44427bbddcbecf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-27 11:13:36.000000000","message":"Patch Set 19:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/27_01_2023_11_05_32\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":19},{"id":"490ad6627ded098dd92b013c69fad0fd712c6941","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-27 12:19:30.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/19/upstream-check/cinder-tempest-plugin-ontap-fcp/9378a7f/ : SUCCESS in 1h 11m 25s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/19/upstream-check/cinder-tempest-plugin-ontap-iscsi/b906066/ : SUCCESS in 43m 42s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/19/upstream-check/cinder-tempest-plugin-ontap-nfs/a7d0c3d/ : SUCCESS in 45m 15s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/19/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/a1a963f/ : FAILURE in 31m 16s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":19},{"id":"f743012f462e94a0e0dcdbf5f8f14c53de9dbe5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 14:38:24.000000000","message":"Patch Set 19: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c63f2f4137e4f1492e3e2436bd06c7e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4500c97572d4dd4a43bf7fcc841c5f2 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f527adc3932b45c2be12178c21a753b2 : SUCCESS in 9m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f44a0cd88d1646569074b5ed5ebcc37b : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f6ea57a77844d6ebab52bc5513251e9 : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/b499c2118c124116a71f083632045d54 : SUCCESS in 1h 02m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9cbddd4c347a460096728a8d8206699b : FAILURE in 1h 14m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0361224979b34a88ab60e58c30c6514e : SUCCESS in 1h 03m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de3a9dce36484d2db0f8c7b20c499e55 : SUCCESS in 8m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6d7e89f05aeb402cbec702a15164ff65 : SUCCESS in 1h 32m 06s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/41454787cdd54a2fbf747bcf28ecbac9 : SUCCESS in 1h 00m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70dcf036431140688e0ac5c3af8cdcb9 : SUCCESS in 41m 36s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b27dbfc270e4c9b835e685036f164c4 : SUCCESS in 1h 07m 11s","accounts_in_message":[],"_revision_number":19},{"id":"e9b6368772a28825372bfe3057255951c29210b3","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-01-27 15:05:03.000000000","message":"Patch Set 19:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/zufens/fc : SUCCESS in 3h 10m 39s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/zufens/iscsi : FAILURE in 3h 16m 21s","accounts_in_message":[],"_revision_number":19},{"id":"8071815a9a66ecd579e92bb43660a63cf60be81a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-01-27 15:25:31.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"05399c42de3b384edfcc051c2866a74a328dff00","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-27 15:28:41.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":19},{"id":"3156c343360ddccafc19376fc3b7bc0438ecf959","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 15:37:43.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5308ea57a9ad4162b95068c499b0f8cb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6027a984d65247b1afd3397be7da9427 : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/47c9c2327eaa475cad0231545d3b14df : SUCCESS in 10m 39s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"60ec23dff2ec06421bd255d7daf2089269944f2e","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-01-27 16:26:42.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/c43243e773124180a6297ad65fc6490d : POST_FAILURE in 23m 29s","accounts_in_message":[],"_revision_number":19},{"id":"e595a88ef185154362070b4f50474e85c85bbedb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 17:37:57.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/736f89d95b4a4eaba1ac73108be5b59c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e1d07b0907f4310b160f0339072e7de : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05a2fb0a2a3c465b86bb4d4e6df53874 : SUCCESS in 10m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38845ad6701043c7adacff6bb729b8d1 : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2efbcebbf2504f03a788a7cd9483a7a0 : SUCCESS in 11m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/d2ee919c8233447d9b0c92b1245dbd77 : SUCCESS in 52m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f2be55d2e358454c913974b6dd36d0aa : SUCCESS in 40m 34s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2298c3a6dc314877b806331435b9585a : SUCCESS in 1h 01m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfcbb4cb375646cb9c3d1ed12db18e15 : SUCCESS in 55m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5e7555e08b54617985a05e913fb9746 : SUCCESS in 7m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/65d26194b2c1424ca578a85e08bfe758 : SUCCESS in 14m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1f78a963232c4da69b179ade1b3dcc07 : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/83e5a4ee51e64adbaa0410d2cfb28fb3 : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c6739189fcb74077935ab20400259866 : SUCCESS in 5m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b064f81df3dd473daf4f6af03e5ae778 : SUCCESS in 5m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6a8efaa4cf1d45ae823d16ef4243c260 : SUCCESS in 55m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aedc767786ad47789a6d4eb2b9212f2f : FAILURE in 4m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e30c38f65b69403ab5b62264f0cfc454 : SUCCESS in 1h 35m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dd4c3f6fb3b240b2b9ff62a0537c9da5 : FAILURE in 1h 23m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0dff2132297b4cd2a111964cca2af419 : SUCCESS in 1h 37m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/edf877eae4f3468dba979c032b045e74 : SUCCESS in 1h 11m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bc85d4e97bae4a2198a1bba9d642ab6a : SUCCESS in 1h 39m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3bb4c5ba5cb746bfa982b69c6eb7f6d7 : SUCCESS in 27m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a9168aa2b2fa4067b879fb87374aff86 : SUCCESS in 1h 39m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cdef44f55e3745da8294678730de0d9a : SUCCESS in 2h 06m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53e0f858e5054d54befb8bda9d410ef7 : SUCCESS in 51m 11s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b540fd2099804bc5b7a489df5adf1c2d : SUCCESS in 1h 20m 04s","accounts_in_message":[],"_revision_number":19},{"id":"e0c8dc26958a9b634d116bddcfaa9246ffad518c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 17:38:45.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"8957c42d930a2d5fe98443e782a1c5277096f7e3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-27 17:58:16.000000000","message":"Patch Set 19:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/27_01_2023_17_50_24\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":19},{"id":"72e3d876fa0d995baf97544f87781f699fdab5e1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-27 18:38:09.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/19/upstream-check/cinder-tempest-plugin-ontap-fcp/265e53e/ : SUCCESS in 59m 14s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/19/upstream-check/cinder-tempest-plugin-ontap-iscsi/f9e6607/ : RETRY_LIMIT in 7m 30s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/19/upstream-check/cinder-tempest-plugin-ontap-nfs/96578c4/ : SUCCESS in 45m 20s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/815582/19/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/6159550/ : SUCCESS in 44m 39s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":19},{"id":"a1ea9154df24e34ae4c0e772b77e8f55e6461b47","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-27 18:52:18.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/94d49c7a909a414fad4eac09832dec70 : SUCCESS in 1h 15m 26s","accounts_in_message":[],"_revision_number":19},{"id":"ef40cc49d70a50867aa90a4658d604414b12e529","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 19:02:54.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89f43404d3f6479d85a392cecb3ea9da\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/607ee2521a7e46fbad5ed5663a301ebc : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c338fcc037694f8a9d7f3a61ab4a0cc7 : SUCCESS in 12m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e53a4ce7522e48d9b398cc60cc127207 : SUCCESS in 10m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2963a910f3694da89a846063dcbb8fde : SUCCESS in 12m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/9e77411a5c374d858e4c3e57925293f0 : SUCCESS in 48m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c71a3de936044a24868884ed67a2b4bb : SUCCESS in 1h 20m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ae992206116491d83f081175136ecbb : SUCCESS in 48m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52510ca0c6ff42e9b5122240799df724 : SUCCESS in 7m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/69515e737008407ba54db6ab15142c1c : SUCCESS in 1h 01m 57s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c93fd143ee9a4f97a08a634e163c7320 : SUCCESS in 1h 01m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1239edc522949d898268992e83a154c : SUCCESS in 44m 20s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1c93178680e46048a3172b58aa07e89 : SUCCESS in 1h 12m 04s","accounts_in_message":[],"_revision_number":19},{"id":"b8c5c75155e4ffd91985df97f1445b7172096237","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 19:02:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":19},{"id":"95a1f2635e03275dc2b171c17b0e11f3333fc249","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 19:04:48.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79185bbe04234f5e9f878b97821ea46b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c62d556ce744ace908598891ccb77d6 : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a682f3225b4d470194722e3a5492dfea : SUCCESS in 57s","accounts_in_message":[],"_revision_number":19},{"id":"5dcfa46f202dd954e2880a2d4d5c473e8f5a0481","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-01-27 22:05:04.000000000","message":"Patch Set 19:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/amvgcs/fc : SUCCESS in 2h 40m 29s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/815582/19/amvgcs/iscsi : FAILURE in 3h 27m 31s","accounts_in_message":[],"_revision_number":19},{"id":"d801aceb515c1d268e655718b51aa0f0b2a296f8","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-01-27 23:02:32.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"4c4ef264acd431d00eb17ceb90fefeb80f07c3f1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-28 06:17:27.000000000","message":"Patch Set 19:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/cf46fb892bd740c48b06ae15099e6fcf : RETRY_LIMIT in 1m 59s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/f975fa6451504bf78b7905b71affc93c : RETRY_LIMIT in 1m 34s","accounts_in_message":[],"_revision_number":19},{"id":"94f122f47594805f85fe46f28e322f80f3dcb437","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-01-28 11:42:47.000000000","message":"Patch Set 19:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1017 \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":19},{"id":"69461d112d0a65ad54744a778041940c5938c576","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-01-28 21:26:53.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/815582-19/HPMSA-iSCSI/ : FAILURE in 1:35:40\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/815582-19/HPMSA-FC/ : SUCCESS in 1:17:55","accounts_in_message":[],"_revision_number":19},{"id":"34d0fa413242a2cb31cde57d858d31d2a247cee5","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-01-29 12:26:46.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/815582-19/HPMSA-iSCSI/ : SUCCESS in 1:36:26\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/815582-19/HPMSA-FC/ : FAILURE in 1:16:29","accounts_in_message":[],"_revision_number":19},{"id":"6de70d018eb1b4842d54e89f06cbc93eb077470c","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-01-30 22:59:58.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/815582-19/PVME-iSCSI/ : SUCCESS in 1:55:23\n- PVME-FC https://os-logs.seagate.com/dell-ci/815582-19/PVME-FC/ : FAILURE in 1:29:55","accounts_in_message":[],"_revision_number":19},{"id":"2f144a78a84c068b0f1574501626c49a31fa4e09","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-01-30 23:07:29.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/815582-19/Seagate-iSCSI/ : SUCCESS in 1:55:25\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/815582-19/Seagate-FC/ : SUCCESS in 1:34:39","accounts_in_message":[],"_revision_number":19},{"id":"b13683ab5713d353f4db15d23290f2d2cf95f784","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-01-31 20:03:52.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/815582-19/PVME-iSCSI/ : SUCCESS in 1:49:39\n- PVME-FC https://os-logs.seagate.com/dell-ci/815582-19/PVME-FC/ : SUCCESS in 1:33:12","accounts_in_message":[],"_revision_number":19},{"id":"3245713ca53d4bdeba782e73110185e27224e0c6","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-01-31 20:17:31.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/815582-19/Seagate-iSCSI/ : SUCCESS in 1:52:07\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/815582-19/Seagate-FC/ : SUCCESS in 1:34:23","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"83fe53081ecab278395b4c3c45ec5a35301a1b01","revisions":{"be27ed612ae38e2a245d33d6e6320975738a0472":{"kind":"REWORK","_number":1,"created":"2021-10-27 05:00:12.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/1"}}},"commit":{"parents":[{"commit":"683744217f43ad84b5567faab2b4a8d2e2f759d4","subject":"Hitachi: Make the parameters name variable for supporting OEM storages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/683744217f43ad84b5567faab2b4a8d2e2f759d4"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2021-10-26 11:12:05.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2021-10-26 11:12:05.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI arrays.\n\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be27ed612ae38e2a245d33d6e6320975738a0472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be27ed612ae38e2a245d33d6e6320975738a0472"}]},"branch":"refs/heads/master"},"d29415b6fc9945a0de937dff0090a99f6fccf1b1":{"kind":"NO_CHANGE","_number":2,"created":"2021-10-27 05:06:08.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/2"}}},"commit":{"parents":[{"commit":"42ea10fe67457d445d6b35c1fe8e02f277a73ba5","subject":"Hitachi: Make the parameters name variable for supporting OEM storages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42ea10fe67457d445d6b35c1fe8e02f277a73ba5"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2021-10-26 11:12:05.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2021-10-27 05:06:08.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI arrays.\n\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d29415b6fc9945a0de937dff0090a99f6fccf1b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d29415b6fc9945a0de937dff0090a99f6fccf1b1"}]},"branch":"refs/heads/master","description":"Rebase"},"7c4ff99bf48dfda90f142686fbe5cf1c5943aaff":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-10-28 11:13:26.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/3"}}},"commit":{"parents":[{"commit":"5c32b29b2720e107f5e187cc04a070272d561bf6","subject":"Hitachi: Make the parameters name variable for supporting OEM storages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c32b29b2720e107f5e187cc04a070272d561bf6"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2021-10-26 11:12:05.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2021-10-28 11:13:26.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI arrays.\n\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c4ff99bf48dfda90f142686fbe5cf1c5943aaff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c4ff99bf48dfda90f142686fbe5cf1c5943aaff"}]},"branch":"refs/heads/master","description":"Rebase"},"463ac005bee86619c22bf5b43754172085d2ae76":{"kind":"REWORK","_number":4,"created":"2021-10-29 00:27:05.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/4"}}},"commit":{"parents":[{"commit":"5c32b29b2720e107f5e187cc04a070272d561bf6","subject":"Hitachi: Make the parameters name variable for supporting OEM storages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c32b29b2720e107f5e187cc04a070272d561bf6"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2021-10-28 11:03:29.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2021-10-28 11:03:29.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI arrays.\n\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/463ac005bee86619c22bf5b43754172085d2ae76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/463ac005bee86619c22bf5b43754172085d2ae76"}]},"branch":"refs/heads/master"},"80bbd34f37a922523813b4450feab19a4e9de228":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-10-29 00:29:58.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/5"}}},"commit":{"parents":[{"commit":"48edd1dd9dc50e372151c7a1c8ef180bfc274119","subject":"Merge \"[Pure Storage] Add missing DB values when performing create CG from CG snap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/48edd1dd9dc50e372151c7a1c8ef180bfc274119"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2021-10-28 11:03:29.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2021-10-29 00:29:58.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI arrays.\n\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/80bbd34f37a922523813b4450feab19a4e9de228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/80bbd34f37a922523813b4450feab19a4e9de228"}]},"branch":"refs/heads/master","description":"Rebase"},"1eb51d3afeb6de79035e60ea603d1d140bb474ca":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-11-25 00:53:13.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/6"}}},"commit":{"parents":[{"commit":"5c32b29b2720e107f5e187cc04a070272d561bf6","subject":"Hitachi: Make the parameters name variable for supporting OEM storages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c32b29b2720e107f5e187cc04a070272d561bf6"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2021-10-28 11:03:29.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2021-11-25 00:53:13.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI arrays.\n\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1eb51d3afeb6de79035e60ea603d1d140bb474ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1eb51d3afeb6de79035e60ea603d1d140bb474ca"}]},"branch":"refs/heads/master","description":"Rebase"},"b047755fbb078260c67e52ba680607c86d7c5c01":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-01-17 08:47:44.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/7"}}},"commit":{"parents":[{"commit":"4082ed3beaf8a3f2787503e8f18119c3d5025138","subject":"Hitachi: Make the parameters name variable for supporting OEM storages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4082ed3beaf8a3f2787503e8f18119c3d5025138"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2021-10-28 11:03:29.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-01-17 08:47:44.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI arrays.\n\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b047755fbb078260c67e52ba680607c86d7c5c01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b047755fbb078260c67e52ba680607c86d7c5c01"}]},"branch":"refs/heads/master","description":"Rebase"},"3053ce597f150350d3313efc61b2f387e8f0970c":{"kind":"REWORK","_number":8,"created":"2022-01-21 08:09:42.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/8"}}},"commit":{"parents":[{"commit":"4082ed3beaf8a3f2787503e8f18119c3d5025138","subject":"Hitachi: Make the parameters name variable for supporting OEM storages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4082ed3beaf8a3f2787503e8f18119c3d5025138"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-01-21 06:08:00.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-01-21 06:08:00.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI arrays.\n\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3053ce597f150350d3313efc61b2f387e8f0970c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3053ce597f150350d3313efc61b2f387e8f0970c"}]},"branch":"refs/heads/master"},"aca8e5be4ab287c8f5cfa7a33bc9bf94fa6ef252":{"kind":"REWORK","_number":9,"created":"2022-06-23 00:46:26.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/9"}}},"commit":{"parents":[{"commit":"e205e008d3f1754d9da2920f3b7a0aa245c59a95","subject":"Hitachi and OEM: Support multi pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e205e008d3f1754d9da2920f3b7a0aa245c59a95"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-06-22 12:15:48.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-06-22 12:15:48.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI arrays.\n\nImplements: blueprint hpe-xp-fc-iscsi-cinder-driver\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aca8e5be4ab287c8f5cfa7a33bc9bf94fa6ef252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aca8e5be4ab287c8f5cfa7a33bc9bf94fa6ef252"}]},"branch":"refs/heads/master"},"f50565252a363cc4595efd3e5841d087f08859e9":{"kind":"REWORK","_number":10,"created":"2022-06-23 08:23:02.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/10"}}},"commit":{"parents":[{"commit":"e205e008d3f1754d9da2920f3b7a0aa245c59a95","subject":"Hitachi and OEM: Support multi pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e205e008d3f1754d9da2920f3b7a0aa245c59a95"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-06-22 12:15:48.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-06-23 08:23:01.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI arrays.\n\nImplements: blueprint hpe-xp-fc-iscsi-cinder-driver\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f50565252a363cc4595efd3e5841d087f08859e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f50565252a363cc4595efd3e5841d087f08859e9"}]},"branch":"refs/heads/master"},"f32cdbe8f97bb11f982cae0482d6e2a936c1e891":{"kind":"REWORK","_number":11,"created":"2022-07-28 00:18:36.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/11"}}},"commit":{"parents":[{"commit":"e205e008d3f1754d9da2920f3b7a0aa245c59a95","subject":"Hitachi and OEM: Support multi pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e205e008d3f1754d9da2920f3b7a0aa245c59a95"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-06-22 12:15:48.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-07-27 07:20:25.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI arrays.\n\nImplements: blueprint hpe-xp-fc-iscsi-cinder-driver\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f32cdbe8f97bb11f982cae0482d6e2a936c1e891"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f32cdbe8f97bb11f982cae0482d6e2a936c1e891"}]},"branch":"refs/heads/master"},"6603b188de81d8ce9f400f52ae5abb710da98589":{"kind":"REWORK","_number":12,"created":"2022-08-05 00:51:52.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/12"}}},"commit":{"parents":[{"commit":"e205e008d3f1754d9da2920f3b7a0aa245c59a95","subject":"Hitachi and OEM: Support multi pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e205e008d3f1754d9da2920f3b7a0aa245c59a95"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-08-05 00:38:01.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-08-05 00:38:01.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI arrays.\n\nImplements: blueprint hpe-xp-fc-iscsi-cinder-driver\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6603b188de81d8ce9f400f52ae5abb710da98589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6603b188de81d8ce9f400f52ae5abb710da98589"}]},"branch":"refs/heads/master"},"8509a870fecc83a03876bd518951e3a0a3d7d6b9":{"kind":"REWORK","_number":13,"created":"2022-08-05 01:07:17.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/13"}}},"commit":{"parents":[{"commit":"e205e008d3f1754d9da2920f3b7a0aa245c59a95","subject":"Hitachi and OEM: Support multi pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e205e008d3f1754d9da2920f3b7a0aa245c59a95"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-08-05 00:38:01.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-08-05 01:07:00.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI arrays.\n\nImplements: blueprint hpe-xp-fc-iscsi-cinder-driver\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8509a870fecc83a03876bd518951e3a0a3d7d6b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8509a870fecc83a03876bd518951e3a0a3d7d6b9"}]},"branch":"refs/heads/master"},"d32422f68cb3b63b93d41c19bb7ba2a4a6ca404a":{"kind":"REWORK","_number":14,"created":"2022-08-05 04:21:38.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/14"}}},"commit":{"parents":[{"commit":"e205e008d3f1754d9da2920f3b7a0aa245c59a95","subject":"Hitachi and OEM: Support multi pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e205e008d3f1754d9da2920f3b7a0aa245c59a95"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-08-05 00:38:01.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-08-05 04:21:37.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI arrays.\n\nImplements: blueprint hpe-xp-fc-iscsi-cinder-driver\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d32422f68cb3b63b93d41c19bb7ba2a4a6ca404a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d32422f68cb3b63b93d41c19bb7ba2a4a6ca404a"}]},"branch":"refs/heads/master"},"147ea5844f0937241c018ea7503f8863f1680808":{"kind":"REWORK","_number":15,"created":"2022-08-05 07:16:11.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/15"}}},"commit":{"parents":[{"commit":"e205e008d3f1754d9da2920f3b7a0aa245c59a95","subject":"Hitachi and OEM: Support multi pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e205e008d3f1754d9da2920f3b7a0aa245c59a95"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-08-05 00:38:01.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-08-05 07:16:11.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI arrays.\n\nImplements: blueprint hpe-xp-fc-iscsi-cinder-driver\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/147ea5844f0937241c018ea7503f8863f1680808"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/147ea5844f0937241c018ea7503f8863f1680808"}]},"branch":"refs/heads/master"},"0c6a82c16f6f20c80700ad42751abf559578a6f0":{"kind":"REWORK","_number":16,"created":"2022-10-17 02:59:45.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/16"}}},"commit":{"parents":[{"commit":"ad184ca698d4b9cce912fcc1e936ef02c44b47d9","subject":"Merge \"NFS: Use Volume attr instead of dict compat for lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad184ca698d4b9cce912fcc1e936ef02c44b47d9"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-10-17 02:45:15.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-10-17 02:45:15.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI\narrays.\n\nImplements: blueprint hpe-xp-fc-iscsi-cinder-driver\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0c6a82c16f6f20c80700ad42751abf559578a6f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0c6a82c16f6f20c80700ad42751abf559578a6f0"}]},"branch":"refs/heads/master"},"11d691bc21c54cc472d87456fc6a473ba145013c":{"kind":"REWORK","_number":17,"created":"2022-10-18 10:51:10.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/17"}}},"commit":{"parents":[{"commit":"ad184ca698d4b9cce912fcc1e936ef02c44b47d9","subject":"Merge \"NFS: Use Volume attr instead of dict compat for lock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad184ca698d4b9cce912fcc1e936ef02c44b47d9"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-10-17 02:45:15.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-10-18 10:51:10.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI\narrays.\n\nImplements: blueprint hpe-xp-fc-iscsi-cinder-driver\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11d691bc21c54cc472d87456fc6a473ba145013c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11d691bc21c54cc472d87456fc6a473ba145013c"}]},"branch":"refs/heads/master"},"3718aee68f1c4ae5be9c10d5ecb9438c890f3642":{"kind":"REWORK","_number":18,"created":"2022-10-26 10:26:02.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/18"}}},"commit":{"parents":[{"commit":"e9482b7f64930624ffcf07ee6581480dbd877ac0","subject":"Hitachi and OEM: Support multi pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9482b7f64930624ffcf07ee6581480dbd877ac0"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-10-26 10:09:26.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-10-26 10:09:26.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI\narrays.\n\nImplements: blueprint hpe-xp-fc-iscsi-cinder-driver\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3718aee68f1c4ae5be9c10d5ecb9438c890f3642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3718aee68f1c4ae5be9c10d5ecb9438c890f3642"}]},"branch":"refs/heads/master"},"83fe53081ecab278395b4c3c45ec5a35301a1b01":{"kind":"REWORK","_number":19,"created":"2022-10-27 00:42:29.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/82/815582/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/82/815582/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/82/815582/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/82/815582/19"}}},"commit":{"parents":[{"commit":"e9482b7f64930624ffcf07ee6581480dbd877ac0","subject":"Hitachi and OEM: Support multi pool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9482b7f64930624ffcf07ee6581480dbd877ac0"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-10-26 10:09:26.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-10-27 00:42:29.000000000","tz":0},"subject":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays","message":"HPE: Cinder driver for HPE XP storage FC and iSCSI arrays\n\nThis patch adds cinder driver for HPE XP storage FC and iSCSI\narrays.\n\nImplements: blueprint hpe-xp-fc-iscsi-cinder-driver\nChange-Id: I6e4ebcb9dc67e444157aaa85fcd0239817a79136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/83fe53081ecab278395b4c3c45ec5a35301a1b01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/83fe53081ecab278395b4c3c45ec5a35301a1b01"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}]}],"submit_requirements":[]}
