)]}'
{"id":"openstack%2Fcinder~815660","triplet_id":"openstack%2Fcinder~master~I09a8357e15b9041790285ab1f9a3cdd7a18243b4","project":"openstack/cinder","branch":"master","topic":"driver-dev-docs-1","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-04-07 19:51:59.000000000","reason":"Brian Rosmaita replied on the change"},"7198":{"account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"last_update":"2022-08-11 13:23:49.000000000","reason":"Change was submitted"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-08-11 13:23:49.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-08-10 10:45:29.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I09a8357e15b9041790285ab1f9a3cdd7a18243b4","subject":"Docs: Document clone_image driver method","status":"MERGED","created":"2021-10-27 13:14:15.000000000","updated":"2022-08-13 10:30:31.000000000","submitted":"2022-08-11 13:23:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":38,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815660-driver-dev-docs-1","meta_rev_id":"8259da3f6559708ea6a83d2d2fc9e190d9c72885","_number":815660,"virtual_id_number":815660,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-11 13:23:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":2,"date":"2022-05-12 21:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2022-08-11 09:20:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2022-05-12 21:51:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-27 13:21:25.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-27 13:44:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-27 14:21:35.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-10-27 14:25:24.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-27 15:37:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-27 16:06:23.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2021-10-27 16:24:18.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-27 16:49:06.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-10-27 16:54:47.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-10-27 17:53:10.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-10-27 19:08:49.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-28 09:09:51.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-28 09:10:03.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-10-28 09:42:02.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-28 11:36:54.000000000","updated_by":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"reviewer":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"state":"CC"},{"updated":"2021-10-28 11:43:21.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-10-28 11:53:20.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-28 12:33:52.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-10-28 12:45:30.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-10-28 15:07:21.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-10-28 17:57:14.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-11-03 14:55:44.000000000","updated_by":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"reviewer":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"state":"REVIEWER"},{"updated":"2021-11-05 08:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2021-11-09 02:06:00.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-12-02 19:18:29.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2021-12-03 01:05:46.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-03 04:49:26.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-03 04:58:11.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-12-04 22:41:26.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-31 19:26:45.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-01-31 20:48:17.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-01-31 22:53:00.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-02-02 02:25:51.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-06 14:31:42.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2022-04-06 14:39:58.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-04-06 20:18:07.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-12 18:32:31.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-05-13 18:01:01.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2022-08-10 13:57:08.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-08-10 15:00:24.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-08-11 02:10:11.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-08-11 09:20:17.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"f2fb110f92f8e27c36cf2c084c31203c11548708","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-10-27 13:14:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2284ad91f3ad6c3bd607e5be1912689f74f53db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-10-27 13:20:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"90aa91465ef40f2165a4c5441c60b242e75bd0b9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-27 13:21:25.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/815660/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"ebac18e985fbac7fef6ec825304b13d3509a4b9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 13:44:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b2772515489480f80ed4da505c968e0 : SUCCESS in 16m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6af2a228738444c8b4c4fd833e4eac75 : SUCCESS in 21m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d9757b1ccdbedb5014ff48466d74e482fd64b930","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-27 13:57:33.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/815660/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"43763bf40d57cbb30090d93eb7a7e527b9ea3bd5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-10-27 14:21:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815660/2/iscsi/ : SUCCESS in 57m 24s\n- cinder-smb http://cloudbase-ci.com/cinder/815660/2/smb/ : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":2},{"id":"413866cb22d693616cb2fb203bf64a91db60ca53","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-27 14:25:24.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k finger://openstackcifc-virtual-machine/e1db0c707218429e840e076ed62af608 : RETRY_LIMIT in 10s","accounts_in_message":[],"_revision_number":2},{"id":"a825f1bada898ba6465abcff212739d1b4cf6f14","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-27 15:05:12.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/815660/2/check/iscsi-driver-primera/67184c2/67184c2.html : FAILURE in 51m 33s","accounts_in_message":[],"_revision_number":2},{"id":"98877cd22de63c1f49b51907049065269db75b11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 15:37:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a225718226fd4099bf8fcfc713a0068d : SUCCESS in 8m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b42c630188e4403b0a1f82233bd0170 : SUCCESS in 10m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dc02b99809e4419be848bc4800d48dd : SUCCESS in 11m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/240b69976aa2449f9802cc29988adb69 : SUCCESS in 13m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf6425808c1640b1b38547925a913433 : SUCCESS in 14m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/1d242525bdfb48638f76f897f9bb73d8 : SUCCESS in 52m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5bf1a6a944b143dca4332f2159a23e64 : SUCCESS in 1h 15m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bb920f7b63bb4b0d87cfbef752001f5d : SUCCESS in 13m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/55c1f7e794e24235af1a4dc5d72454a5 : SUCCESS in 8m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a9e7980481bc4e03b199011a0a29a406 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/87b39db9d314481181b8b8639536456a : SUCCESS in 6m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c87e6411c55449f89251e6ff3dc2a2d0 : SUCCESS in 6m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0b4ba55ee8984a93b35fc709e1ef7a34 : SUCCESS in 44m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b19748321b0e4e6a8fa25307d24563bf : FAILURE in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/44c75eea68c1400e84092a4921ce2a4f : SUCCESS in 1h 21m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/62bbbead494844e29379c71755d17fed : SUCCESS in 1h 26m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a55ef2732ca84d3487a269a5cc8f5c40 : FAILURE in 1h 17m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/342011b9eb8a4c08ba7102225b84f536 : SUCCESS in 1h 05m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c82fc647c0594d45bebca940f27e0e30 : SUCCESS in 1h 16m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5ec7415042364cc0954b19c62382088c : SUCCESS in 47m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/694cb0c7fd424e77ae82177e4b35479c : SUCCESS in 1h 20m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/55d42cc5dce440f1a6229b705e6750e3 : SUCCESS in 2h 13m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87a410644fe449609b986d69504879ef : SUCCESS in 1h 04m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/439dea4bc42344a2ba7703ea36a07e00 : SUCCESS in 48m 44s","accounts_in_message":[],"_revision_number":2},{"id":"0d1f51a7d5a8b776b9f405e983136d9b8ff163a1","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-10-27 16:06:23.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"304a633c4280b4767d28b30216ccb30b1ecf8157","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-10-27 16:24:18.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/60/815660/2/check/DellEMC_XtremIO_FC/d00729d/DellEMC_XtremIO_FC/55 : SUCCESS in 1h 03m 13s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/2/check/DellEMC_XtremIO_iSCSI/9b73bad/DellEMC_XtremIO_iSCSI/53 : FAILURE in 27m 34s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"253c73232a7317d52526fc0127890b86d3692e79","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-27 16:49:06.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/815660/2/check/DellEMC_Unity_FC/0c24382/DellEMC_Unity_FC/95 : FAILURE in 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/2/check/DellEMC_Unity_iSCSI/0c7c429/DellEMC_Unity_iSCSI/90 : FAILURE in 1h 38m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"449cb1a7f30c57e0f00a7bcb510c5c8d81c168f7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-27 16:49:18.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/815660/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"31e3371bf608c57cc824f867e497ef798a1576b0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-10-27 16:54:47.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/815660/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/fab89a4/ : SUCCESS in 2h 34m 57s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/815660/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/048e699/ : TIMED_OUT in 3h 06m 23s","accounts_in_message":[],"_revision_number":2},{"id":"5f95c6455ae126c84e2fbf3983cbae523188d8b5","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-10-27 17:01:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a7cb595ca48bde5a7c44b0f41668a3c1cb0adf70","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-27 17:13:59.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/815660/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"8715e6eed8d861e6dc780c29a1d11b0640300de9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-10-27 17:53:10.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/815660/2/check/dsvm-tempest-cisco-zonemanager-job/787b44e/ : FAILURE in 59m 54s","accounts_in_message":[],"_revision_number":2},{"id":"231425d5f0c1e7981e2886a8005e73cd50058ed4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-27 17:58:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/815660/2/check/iscsi-driver-nimble/7f31f01/7f31f01.html : FAILURE in 58m 07s","accounts_in_message":[],"_revision_number":2},{"id":"829bb67947615e5ca164e7c344f2ef7dae63bb0c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-27 19:08:49.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-815660-1/ns5_nfs : SUCCESS in 4h 39m 31s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-815660-1/ns5_iscsi : SUCCESS in 4h 40m 37s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"0918f5d154e40c940a7c59790ffe75092baa8dee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-27 19:46:48.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/815660/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"70fe50d2dc5c5ef195a9a689fcc637d8f43993a9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-27 21:00:31.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-815660-2/ns5_nfs : SUCCESS in 4h 36m 25s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-815660-2/ns5_iscsi : SUCCESS in 4h 45m 37s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"9d6eb5bf2190bb3541363dd635e5cb71d796b4c0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-10-27 21:20:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/815660/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/26586d2/ : SUCCESS in 2h 29m 43s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/815660/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b6f7248/ : SUCCESS in 1h 41m 36s","accounts_in_message":[],"_revision_number":2},{"id":"89141a5d0da7034a2153b273ab073b95ca2877b9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-27 21:23:23.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/815660/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"5b2cf1d4a8dcc6580d41f0e5af6a1241f31c991a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-10-28 07:53:49.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c181358115c65a6b2d5e4e0ffae1369bfe3fb6da","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-28 08:16:30.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/815660/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"4069ad418907ce1276daf174d04b3417715e8a7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 08:27:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/300e33954a074299adda948a8a9df961 : SUCCESS in 14m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75aac478f17e4adcb7d00b44f62e0827 : SUCCESS in 20m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b5a9e13483afb02c3ef92e6b7b8cd499bafc1943","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-28 08:35:36.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/60/815660/3/check/iscsi-driver-alletra-6k/9fd0494/9fd0494.html : FAILURE in 38m 09s","accounts_in_message":[],"_revision_number":3},{"id":"40b3e18272077cf7d67795f10b822e7d6e0b74f3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-28 08:39:10.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/815660/3/check/DellEMC_Unity_FC/34ff4d8/DellEMC_Unity_FC/112 : FAILURE in 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/3/check/DellEMC_Unity_iSCSI/6e0f677/DellEMC_Unity_iSCSI/107 : FAILURE in 21m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"c4eac3a6db752701a3b6ac34e08d75ea0b4e3ba7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-28 08:53:52.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/815660/3/check/iscsi-driver-primera/8d13e2f/8d13e2f.html : FAILURE in 55m 28s","accounts_in_message":[],"_revision_number":3},{"id":"b06ecdf8d58bbbdc5fb65cbd0c3ac87f446abb9c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-10-28 08:59:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815660/3/iscsi/ : SUCCESS in 1h 00m 41s\n- cinder-smb http://cloudbase-ci.com/cinder/815660/3/smb/ : SUCCESS in 1h 04m 55s","accounts_in_message":[],"_revision_number":3},{"id":"8e64825abb840d01c38bbc778da9310b6ef91cf8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-28 09:02:19.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/815660/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"36f07f51bb17a7795d3d944b975d2510bbb3b693","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-10-28 09:09:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/2/20211028151328/NEC-ISCSI : SUCCESS in 2h 47m 47s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/2/20211028151328/NEC-FC : SUCCESS in 2h 56m 20s","accounts_in_message":[],"_revision_number":2},{"id":"05a7a29d4917003de59d65910fdcd3f7e5f408a6","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-10-28 09:10:03.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"997d23eeb75639bb322284faa4f659e7f98a0de4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-10-28 09:42:02.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/815660/3/check/tempest-dsvm-full-focal-py3/22b48b8/ : FAILURE in 1h 31m 23s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/815660/3/check/devstack-plugin-ceph-dsvm-focal-py3/2889f11/ : FAILURE in 1h 34m 40s","accounts_in_message":[],"_revision_number":3},{"id":"c6b76e02069d0969b25afdffbcbcef66e36375f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 10:06:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/036c2b1dd35847eca903c1322b1c072a : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0798a9b4982b498389a0bd8d4edb4c65 : SUCCESS in 10m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/143607161dce4d3e990f4f5402292159 : SUCCESS in 11m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/598298eeec0d4ce9a0bba8f4d8d0c4ce : SUCCESS in 12m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bf33285fa514900a8620ece297bce70 : SUCCESS in 13m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/f2a1009e7f204743884390fc1f041701 : SUCCESS in 38m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0e99efaa62f149d0aa7d66bbc2e68be9 : SUCCESS in 1h 23m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7f9455176639411a8e445c30ea2b0f47 : SUCCESS in 16m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e978bff5f966492fbfe9dc87b8d35801 : SUCCESS in 7m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4bf4a5d8a3ff43959231d6a081eda3bd : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/99de2404ab9d41d6ac5764dc56c6f906 : SUCCESS in 8m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/04e7b0b364774cbf98a96d5746a20ac1 : SUCCESS in 7m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/87a9104b647f49fd9b19c379b1c4fcd1 : SUCCESS in 41m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8c52c27e316949b5bd3fa55eff6eea04 : FAILURE in 6m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8a012a7a84fa4c4294ec316bffb099be : SUCCESS in 1h 11m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/62bce9400d844a54a89d6746cab37bee : SUCCESS in 1h 31m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cd812b115d0e492590a0b6885f2bb160 : SUCCESS in 1h 28m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b643f291b843477fa55393d10d3d2ecf : SUCCESS in 1h 20m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9eab37a817b840a389e271e6add402e5 : SUCCESS in 1h 14m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5e260536b96549f88730fe1c21da65d5 : SUCCESS in 39m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6644a830ab9a4aecbb25e667b6ed4478 : SUCCESS in 1h 18m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ad168d2ee67443b9ee16f5f942db691 : SUCCESS in 2h 08m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97f437c54c794e80bd2785be02be9a9a : SUCCESS in 53m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59c3233f9a46433ca3fa20e9e292b7f0 : SUCCESS in 47m 49s","accounts_in_message":[],"_revision_number":3},{"id":"1cf67fbd964ef840edb8585a70b6c8d7cc748a66","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-28 10:08:27.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/815660/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"699952c7b4cae7d97c2593e41ac3350a38ca40f6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-10-28 10:40:00.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/60/815660/3/check/dsvm-tempest-cisco-zonemanager-job/f971475/ : FAILURE in 1h 00m 11s","accounts_in_message":[],"_revision_number":3},{"id":"ee42a0706343f2411549244605f29bff339ddcbd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-10-28 10:58:28.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/815660/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b3673c2/ : TIMED_OUT in 3h 02m 27s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/815660/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3c5f8c1/ : TIMED_OUT in 3h 02m 36s","accounts_in_message":[],"_revision_number":3},{"id":"4d7ae63c0cd282a37c9a223ee8912713bd5f595d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2021-10-28 11:36:54.000000000","message":"Patch Set 1:\n\nBuild Failed. ibm-gpfs-dsvm-tempest-full: s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/64/index.html","accounts_in_message":[],"_revision_number":1},{"id":"d5aceb839ccb36f29a294b7362ed762934fdf44d","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-10-28 11:43:21.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/60/815660/3/upstream-solidfire-check/cinder-solidfire-iSCSI/8cd546a/ : FAILURE in 1h 22m 53s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"dde87476cf50a96c04eb811c43b6c1571aa0289e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-10-28 11:51:27.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/815660/3/check/iscsi-driver-nimble/8af5dec/8af5dec.html : FAILURE in 59m 02s","accounts_in_message":[],"_revision_number":3},{"id":"064f5be25c8348a74ddc4cba68c3098d3ed8d82a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-10-28 11:53:20.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/60/815660/3/check/DellEMC_PowerFlex/48b7777/DellEMC_PowerFlex/46 : SUCCESS in 1h 46m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"38931649c1ff3b65f0eb86fc181e5571d2fdd7d3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-28 12:14:42.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-815660-3/ns5_nfs : SUCCESS in 4h 16m 38s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-815660-3/ns5_iscsi : SUCCESS in 4h 20m 13s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"bc35c740eeff1611222204c54dda35b5afbd85f9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-28 12:14:46.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/815660/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"2d6f81b160b9185989844d71ef93449b5e267a18","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-10-28 12:33:52.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI http://os-logs.tristero.net/dell-ci/815660-3/PVME-iSCSI : SUCCESS in 56:40.17\n- PVME-FC http://os-logs.tristero.net/dell-ci/815660-3/PVME-FC : SUCCESS in 59:49.41","accounts_in_message":[],"_revision_number":3},{"id":"1c72e398c00072d3ebb089edad97cfa9f91a2a0f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-10-28 12:45:27.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/815660/3/check/tempest-dsvm-full-focal-py3/a5890a2/ : FAILURE in 1h 29m 12s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/815660/3/check/devstack-plugin-ceph-dsvm-focal-py3/adc89d6/ : FAILURE in 1h 15m 17s","accounts_in_message":[],"_revision_number":3},{"id":"1214f092314a798e77a25321a009666778284b87","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-10-28 12:45:30.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/815660-3/Seagate-iSCSI : SUCCESS in 1:20:02\n- Seagate-FC http://os-logs.tristero.net/seagate-ci/815660-3/Seagate-FC : SUCCESS in 48:57.50","accounts_in_message":[],"_revision_number":3},{"id":"7c4c29cd87119a22aba76c4274d52d435c97b080","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-28 12:45:50.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/815660/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"451d2c200fb7c21ca5dcbd536e25e85d8e8d3655","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-28 13:02:04.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/815660/3/check/DellEMC_Unity_FC/6f7d329/DellEMC_Unity_FC/116 : FAILURE in 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/3/check/DellEMC_Unity_iSCSI/53b7b17/DellEMC_Unity_iSCSI/111 : FAILURE in 1h 34m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"3207023b632eb3361f2f21f693c99f6174482ac8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-28 13:02:46.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/815660/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"a4c57454a77a24590fd164a1c660b471b1060209","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-28 13:24:04.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/815660/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"66574675332e3cf349f12b2f44b7afc6cc427a79","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2021-10-28 13:44:39.000000000","message":"Patch Set 2:\n\nBuild Failed. ibm-gpfs-dsvm-tempest-full: s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/65/index.html","accounts_in_message":[],"_revision_number":2},{"id":"1cad51d57f7fa9fff9656634b0c727fccf48f2c8","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-10-28 15:07:21.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/815660-3/HPMSA-iSCSI : SUCCESS in 1:08:49\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/815660-3/HPMSA-FC : SUCCESS in 44:19.70","accounts_in_message":[],"_revision_number":3},{"id":"3c1392685d902af72fcadb33f5b7e05c328477fc","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-10-28 17:57:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/815660-3/Lenovo-iSCSI : SUCCESS in 1:23:42\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/815660-3/Lenovo-FC : SUCCESS in 57:24.08","accounts_in_message":[],"_revision_number":3},{"id":"ca55bf6ecc776dc18bcf138cd869bbc8cb9f917e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-10-29 05:10:53.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/3/20211029111858/NEC-FC : SUCCESS in 2h 49m 35s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/3/20211029111858/NEC-ISCSI : SUCCESS in 2h 51m 52s","accounts_in_message":[],"_revision_number":3},{"id":"824ce6bde5a744981ec70d095d4086a22d460e0d","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-11-03 14:55:44.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ece3119cc7777d23092e1cc419bb88afc1266c9b","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2021-11-05 08:00:08.000000000","message":"Patch Set 3:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/815660/3/fc : FAILURE in 1h 47m 16s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/60/815660/3/iscsi : SUCCESS in 1h 19m 46s","accounts_in_message":[],"_revision_number":3},{"id":"676ca7c1dd135ce9ceff4adf889c4c6f975a908d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-11-09 02:06:00.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f5e8b92af23de250f10c753b08ca5070580d44de","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-12-02 18:33:06.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2789fca804a06f0254152fcc80fb6b0fc4657fb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-12-02 18:33:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"49860de3f95c35f94b842a659d0eaecd87758442","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-12-02 18:34:57.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"501861c630dbf5159cf9608a0c9971b4de11c903","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-02 18:48:23.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815660/4/iscsi/ : FAILURE in 13m 50s\n- cinder-smb http://cloudbase-ci.com/cinder/815660/4/smb/ : FAILURE in 13m 48s","accounts_in_message":[],"_revision_number":4},{"id":"bde0fd88ca151c0a4ae41cf4a7d3c2626c009468","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-02 19:18:29.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k finger://openstackcifc-virtual-machine/fd30a1cd7af4443e9f98b9b67b94f136 : RETRY_LIMIT in 10s","accounts_in_message":[],"_revision_number":4},{"id":"c65deb7455da1f46cdacf4b1852fe426af1acb0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 19:20:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37b98f2cd04b49c7ac82c19b424e0a2b : SUCCESS in 18m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7cf6285d09904e0a860a7557cf17ab81 : FAILURE in 23m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5720b5b45bbb94d85b9198f4b4f2092effa9fb6b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-02 21:11:27.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/815660/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"9308f17e8d775a1e3d308c922973330f4dbadc3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 21:16:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d58cad002948b6a46741dcd9aff4a7 : SUCCESS in 8m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c52e52dd38ba44c8928372c601bc55d6 : SUCCESS in 10m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fdec2f4acac40d9a96056d8f7bafe61 : SUCCESS in 11m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eb32bf80366412687adbcd9e7cea715 : SUCCESS in 13m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/fa73718397d94c58b60f59a65065ca83 : SUCCESS in 1h 05m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3f8e21fff4224b1cb61559e5aedb31ba : SUCCESS in 1h 17m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7d47765fcbd45fe9c5cbc22abc66668 : SUCCESS in 47m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/411c9a0cd1a146b3b9682416a5c230bc : SUCCESS in 15m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cd25f8e7c07c4e85a726ef9ea92ae35c : SUCCESS in 6m 28s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4c6df98406ee426eb24953f45ad7d775 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c0bb7d716dea4fbea6e57ab35dd2956a : SUCCESS in 7m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7b7c4dab14f740979368c919ea34b283 : SUCCESS in 7m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0c98a21a13214391881de76c86e4c263 : SUCCESS in 43m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/755e8de7c615421a9b6a8d02e6cef18f : FAILURE in 6m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b0f9da63cedc4e208e5e5cbbfb63acd9 : SUCCESS in 1h 19m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9b4eb7ffb023485695789b8f264a2cb8 : SUCCESS in 1h 30m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d932d489c2294f2eacf5aced4bb8a4b2 : SUCCESS in 1h 37m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9424607bc6c841bfa830283320c07514 : SUCCESS in 1h 05m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/100b1acfc354441fb4a7f6051d8a90db : SUCCESS in 1h 26m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/533812e67e134a44a77070dbe8b713d7 : SUCCESS in 44m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f70b9ef53d414a9bbdcdad2c0e994f4c : SUCCESS in 1h 31m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d0c6e87807384828baa8ce156dc8c564 : SUCCESS in 2h 17m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7462d6ad3f84e0cb1d9a16703a3508d : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":4},{"id":"b6d274e4c5d84d8b4a8afd1ba805543ea3d0f777","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-12-02 21:18:00.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/60/815660/4/check/DellEMC_PowerFlex/c08469f/DellEMC_PowerFlex/183 : FAILURE in 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"e5cf0e6293a5b013434a815ed6da041bd57d26f5","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-12-02 21:57:09.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/60/815660/4/upstream-solidfire-check/cinder-solidfire-iSCSI/d0147fe/ : FAILURE in 38m 12s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"f01360a70e21ecef2513cbb83889adb7d77551ae","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-02 22:10:18.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/815660/4/check/tempest-dsvm-full-focal-py3/391bb3f/ : SUCCESS in 3h 09m 23s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/815660/4/check/devstack-plugin-ceph-dsvm-focal-py3/2fe2a1b/ : TIMED_OUT in 3h 11m 43s","accounts_in_message":[],"_revision_number":4},{"id":"52fcf8cc1c6305038167117d0cd83264cd152c10","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-02 23:55:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/815660/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a133574/ : SUCCESS in 1h 27m 38s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/815660/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a433a1d/ : SUCCESS in 1h 13m 14s","accounts_in_message":[],"_revision_number":4},{"id":"2980fd8c35b27a4b8fd49723ada6650623619e03","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-03 00:29:14.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/815660/4/check/DellEMC_Unity_FC/71f1c1d/DellEMC_Unity_FC/260 : FAILURE in 1h 06m 49s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/4/check/DellEMC_Unity_iSCSI/9fe6344/DellEMC_Unity_iSCSI/262 : SUCCESS in 1h 27m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"ba8708638fc5c033ac0060c720bf6984a546736a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-12-03 01:05:46.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/60/815660/4/upstream-check/cinder-cDOT-iSCSI/8c1048d/ : FAILURE in 1h 59m 17s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/60/815660/4/upstream-check/cinder-cDOT-NFS/c8e7d60/ : FAILURE in 1h 51m 58s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/60/815660/4/upstream-check/cinder-cDOT-FCP/c7bf3ca/ : FAILURE in 3h 47m 28s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"9c829f24bbc10f792cc8112e74e69349478ad3ff","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-03 01:52:31.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-815660-4/ns5_nfs : SUCCESS in 4h 29m 54s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-815660-4/ns5_iscsi : SUCCESS in 4h 22m 18s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"3eaca44c0a5bf8ed88578ee4017962330102aa08","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-03 04:49:26.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/11c648abebed445e87a6dc598817abca : FAILURE in 1h 02m 32s","accounts_in_message":[],"_revision_number":4},{"id":"0c254772135abcce550d21f40c01d3e79cbbf9c8","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-12-03 04:58:11.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/60/815660/4/check/DellEMC_SC_FC/0284922/DellEMC_SC_FC/105 : SUCCESS in 47m 17s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/4/check/DellEMC_SC_iSCSI/a821fb2/DellEMC_SC_iSCSI/104 : FAILURE in 1h 56m 47s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"89f07e5ec281f1791e137850b2e381f71ef15658","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-03 05:19:34.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/815660/4/check/iscsi-driver-nimble/db51448/db51448.html : FAILURE in 57m 56s","accounts_in_message":[],"_revision_number":4},{"id":"d7403b9c8f6bd363fa9187e2f7a71f0f28acaa22","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-03 08:24:56.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/60/815660/4/check/DellEMC_XtremIO_FC/5b3436b/DellEMC_XtremIO_FC/292 : SUCCESS in 1h 06m 13s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/4/check/DellEMC_XtremIO_iSCSI/bb05d0e/DellEMC_XtremIO_iSCSI/307 : SUCCESS in 1h 38m 04s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"ea7bb1adf79ab3d512474dc9400b722620bfd43d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-03 11:01:57.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/815660/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"d2d9a4bf5765f775971345e9096f14aff5c037ed","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-12-03 12:02:46.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/815660-4/HPMSA-iSCSI : SUCCESS in 1:11:29\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/815660-4/HPMSA-FC : SUCCESS in 43:52.08","accounts_in_message":[],"_revision_number":4},{"id":"7b4d8634b3643ab31c829e000cb4d97f7a009d3f","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-12-04 22:41:26.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/60/815660/4/check/DellEMC_PowerMAX_FC/8ed3009/DellEMC_PowerMAX_FC/113 : FAILURE in 3h 39m 08s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/4/check/DellEMC_PowerMAX_iSCSI/b982a43/DellEMC_PowerMAX_iSCSI/111 : FAILURE in 3h 29m 00s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"089e00fd548253edb0f45e23a0f42484c709ed51","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-12-05 02:13:31.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/4/20211205081918/NEC-FC : SUCCESS in 2h 43m 24s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/4/20211205081918/NEC-ISCSI : SUCCESS in 2h 54m 10s","accounts_in_message":[],"_revision_number":4},{"id":"463239c7298aef5a5fa56efc1dd4fe5e4bc3ca72","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-05 02:19:34.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/815660/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"01d7adb841474c7956898c65ab5401a88c0cf2fa","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-05 17:21:24.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/60/815660/4/check/dsvm-tempest-TOYOU-ACS5000-driver/2fe554e : FAILURE in 5h 00m 16s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/60/815660/4/check/fc-test/1c3e64b : FAILURE in 1h 17m 36s","accounts_in_message":[],"_revision_number":4},{"id":"f175be489cd497923043e345d28481bd643cc5da","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-05 19:53:19.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/60/815660/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"8766d71c977e14d08db37f601d40decb8a59f75f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-01-31 15:59:44.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"7f1413999d9557cb084d5f353f8983b2badcc58d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-01-31 16:16:58.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"88a5e1172296a89cfff5f1b78457aa3818f1d793","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-01-31 17:18:00.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/60/815660/5/check/iscsi-driver-alletra-6k/1a25eda/1a25eda.html : FAILURE in 21m 48s","accounts_in_message":[],"_revision_number":5},{"id":"9133d74194103991f047055a3939dbc65f9843cc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-31 17:25:33.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815660/5/iscsi/ : SUCCESS in 1h 04m 55s\n- cinder-smb http://cloudbase-ci.com/cinder/815660/5/smb/ : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":5},{"id":"9dbd638e6dcfe6910d870d95ab3cdd5aca707852","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 17:43:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/20e91c4921af491d84e46411c2963281 : SUCCESS in 21m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/200868f0ef954941ac8bcd4227ae6963 : SUCCESS in 23m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b727bbf2642c79e9d1b1636bb4d7ce5031ba1218","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-01-31 17:55:56.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/815660/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/99c60b0/ : SUCCESS in 1h 49m 32s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/815660/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/e99328f/ : SUCCESS in 56m 50s","accounts_in_message":[],"_revision_number":5},{"id":"530833c6593aebdeeead6c0e84842606e2ea7205","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-01-31 18:37:42.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/815660/5/check/iscsi-driver-primera/949ce5e/949ce5e.html : FAILURE in 52m 27s","accounts_in_message":[],"_revision_number":5},{"id":"f4dd7c8861d78ff18d98c57aef461f7a5a0deb75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 19:18:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fd23f851f8949adb732e63c560ba90d : SUCCESS in 9m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a611e520a35e478d91ed1d4bbd8e0cef : SUCCESS in 12m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16244967fe9b4b6583589f72b517a7fa : SUCCESS in 13m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23f47dc4dd5e43568991325a1a3298cc : SUCCESS in 16m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/324604b2454e411fad35ea08c73a65d0 : SUCCESS in 1h 00m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/970fe472c3b24bbc93a570064fe81bbd : SUCCESS in 58m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/439ead5fccbc49c18c45abe6c63eeefa : SUCCESS in 48m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/27ffeddfdc42443b93666a51863d6d3a : SUCCESS in 15m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9faf9efc57764957bfa79cf281989e53 : SUCCESS in 9m 37s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/86dbc993a25a46eeb07573bed7371dae : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/54fabaf431f941eb815692286be3ab71 : SUCCESS in 10m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/47fa0bc02b8c47159956503be7657fa6 : SUCCESS in 10m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d89ea510026543bc950e806edb4c66c7 : SUCCESS in 36m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4b6e914f9cc54010a49dc7076de454fd : FAILURE in 8m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cef799c5fedc4bb5a24e8f453bfc791d : SUCCESS in 1h 15m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/85cd6cefec794544ad14971ec8fc06c3 : SUCCESS in 1h 46m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1f378153ed074c40bc842cca5e12947d : SUCCESS in 1h 28m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6ae88539f8c6466eab9897ca7c7aeb8e : SUCCESS in 1h 09m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1b06d900d2d84b6fa207c401b2ea272c : FAILURE in 1h 22m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a04b7ceb0fa840199cffcabcdd088457 : SUCCESS in 42m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/430a18616f1548128ba48c80abb434b5 : SUCCESS in 1h 08m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4cee70ea368b4178b0cf057282ba61f9 : SUCCESS in 1h 58m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78263faf874d4343b1c9080dff4a9510 : SUCCESS in 52m 39s","accounts_in_message":[],"_revision_number":5},{"id":"77040fe270ef0bb7b22f61ad44a48657d38d43f3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-31 19:26:45.000000000","message":"Patch Set 5:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/31_01_2022_19:19:25\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"1ac7bace83899e39b288256cfb7cd34463d65659","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-31 19:32:27.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/60/815660/5/check/DellEMC_PowerFlex/8d1eada/DellEMC_PowerFlex/338 : FAILURE in 6m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"953281af0189457615b429bf6091396096991d2f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-31 20:07:49.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/815660/5/check/DellEMC_Unity_FC/ea80d03/DellEMC_Unity_FC/488 : FAILURE in 1h 24m 37s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/5/check/DellEMC_Unity_iSCSI/f82ed43/DellEMC_Unity_iSCSI/482 : SUCCESS in 1h 18m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"f7f2a7230b17ff8b8ede78bf2783a88cfecfdcbf","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-01-31 20:10:00.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/815660/5/check/iscsi-driver-nimble/cd13989/cd13989.html : FAILURE in 1h 01m 01s","accounts_in_message":[],"_revision_number":5},{"id":"5dd214277ff858ad6c6ac6543fd5d2dfbd83b959","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-01-31 20:21:30.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/60/815660/5/check/dsvm-tempest-cisco-zonemanager-job/c6e83cc/ : FAILURE in 1h 11m 09s","accounts_in_message":[],"_revision_number":5},{"id":"ee11d14ae10536190fe19cae407b1c9ad28c778c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-01-31 20:28:56.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/5/upstream-solidfire-check/cinder-solidfire-iSCSI/a0af838/ : SUCCESS in 1h 07m 31s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"b383c8c28fc5bc5551b1bbb119ac9171a91606f3","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-01-31 20:48:17.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-60-815660-5/1643656711 : SUCCESS in 33m 23s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":5},{"id":"bece821022fa6f97772821b4aa564ff6a98f7ba3","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-31 21:10:45.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/5/upstream-check/cinder-cDOT-iSCSI/906d622/ : SUCCESS in 37m 20s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/5/upstream-check/cinder-cDOT-NFS/cfe0195/ : SUCCESS in 39m 12s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/5/upstream-check/cinder-cDOT-FCP/bd3cee1/ : SUCCESS in 1h 50m 48s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"d593a3b79871fb18a183a489f342ed08104ee520","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-31 21:46:04.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/815660/5/check/tempest-dsvm-full-focal-py3/9fc1810/ : SUCCESS in 2h 04m 07s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/815660/5/check/devstack-plugin-ceph-dsvm-focal-py3/ee2f0ac/ : SUCCESS in 2h 29m 55s","accounts_in_message":[],"_revision_number":5},{"id":"2a3cd85ba897750b6522e09f2b5e3ca687de49b9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-31 22:02:45.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-815660-5/ns5_nfs : FAILURE in 53m 40s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-815660-5/ns5_iscsi : SUCCESS in 4h 32m 20s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"d9a0f92cfb2f8d0464b5767550ab9d4437f7e8d1","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-01-31 22:28:02.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/60/815660/5/check/DellEMC_PowerMAX_FC/fd08299/DellEMC_PowerMAX_FC/252 : FAILURE in 53m 12s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/5/check/DellEMC_PowerMAX_iSCSI/836ffae/DellEMC_PowerMAX_iSCSI/265 : FAILURE in 50m 19s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"d657cc3724180af58c6e1c0d51f5cdd63bab9ed2","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-01-31 22:48:13.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/60/815660/5/check/DellEMC_XtremIO_FC/5a364a3/DellEMC_XtremIO_FC/602 : SUCCESS in 59m 21s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/5/check/DellEMC_XtremIO_iSCSI/4eeb498/DellEMC_XtremIO_iSCSI/612 : SUCCESS in 1h 22m 01s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"cefe63d1667740a98e9ce8ba780a72efabbdf6ed","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-01-31 22:53:00.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/5/20220201053555/NEC-V-FC : SUCCESS in 2h 06m 57s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/5/20220201053555/NEC-V-ISCSI : SUCCESS in 2h 17m 02s","accounts_in_message":[],"_revision_number":5},{"id":"1f368dab60910f0fe5a11048391bf83409e384de","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-02-01 01:27:10.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/60/815660/5/check/DellEMC_SC_FC/98860ca/DellEMC_SC_FC/275 : FAILURE in 1h 12m 19s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/5/check/DellEMC_SC_iSCSI/8a52fb1/DellEMC_SC_iSCSI/271 : SUCCESS in 1h 42m 51s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"48992d05a8410bf43ef107d312cadcd6c9d37743","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-01 02:29:20.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/815660-5/HPMSA-iSCSI/ : SUCCESS in 51:38.06\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/815660-5/HPMSA-FC/ : SUCCESS in 42:42.40","accounts_in_message":[],"_revision_number":5},{"id":"515775224ee3201f257a9e927f51cf1dbebc68bd","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-01 04:11:04.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/815660-5/Lenovo-iSCSI/ : SUCCESS in 1:01:41\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/815660-5/Lenovo-FC/ : SUCCESS in 56:24.90","accounts_in_message":[],"_revision_number":5},{"id":"8263755228507b1a7262dc8658afd81915d1e9a2","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-01 04:44:20.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/815660-5/PVME-iSCSI/ : FAILURE in 1:12:59\n- PVME-FC https://os-logs.seagate.com/dell-ci/815660-5/PVME-FC/ : SUCCESS in 1:01:43","accounts_in_message":[],"_revision_number":5},{"id":"0d0ddceb5deee7ff0d022d2fb1ea0c426bc114fd","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-02-01 04:57:43.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/815660-5/Seagate-iSCSI/ : SUCCESS in 1:04:58\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/815660-5/Seagate-FC/ : SUCCESS in 1:05:05","accounts_in_message":[],"_revision_number":5},{"id":"c20e6c8e43a03e01c8d2e1c774972ee35b01b28d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-02 02:25:51.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/815660/5/ : SUCCESS in 1h 41m 56s","accounts_in_message":[],"_revision_number":5},{"id":"2bc68e02b07a97e3e5365ee7a81818ef65695c99","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-02-03 09:23:40.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/5/20220203115929/NEC-ISCSI : SUCCESS in 3h 17m 31s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/5/20220203115929/NEC-FC : SUCCESS in 6h 24m 08s","accounts_in_message":[],"_revision_number":5},{"id":"090ced22735438b0b7f382b4db12b978768d8e86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-04-06 10:11:15.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"7be949965865a7ea9ec0cc6f1ec35691b97b348d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-06 10:52:05.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-60-815660-6/ns5_nfs : FAILURE in 35m 51s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-60-815660-6/ns5_iscsi : FAILURE in 35m 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"9fd7e690757ed5d841b975792bd83193102e37ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 12:11:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f696e553925a4be1bd719eb13b84fcd0 : FAILURE in 30m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b1b404aaeb384efd8acaeb39819bc0da : FAILURE in 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"dc89ea20a653e5721032c293b4f9e803a06aa6fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 12:41:57.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0d0e70a61f84a32a199647fd0a9ce6e : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96ae94a4edb24b99ad02001c65d78594 : FAILURE in 11m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d763e74e0124fa2b8f448dfa6455bd4 : FAILURE in 12m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8875c7d041a84ef08c6d81dacb8005e4 : RETRY_LIMIT in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c26274b3e404ee3910b1a3b3231e874 : SUCCESS in 12m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/aab007cb73644e218a496f1c320d3010 : SUCCESS in 56m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b2ac4839f5bf44be93d61de0462ce128 : SUCCESS in 1h 14m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/532991420ae74483a4ab4585235ff186 : SUCCESS in 56m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/18a16150a32444bea64985a9a40d3652 : FAILURE in 13m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a65210274d4d4c82a95a7a998390d7a0 : SUCCESS in 6m 40s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/11a6f85f63144e7883776c9b81ebe738 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3081d7d06a4449979fdd15cec8abcb7c : SUCCESS in 6m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d67d155ed8ed4577ac406074a546658b : SUCCESS in 8m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d45819e12bf444d7985115e98ae27a37 : SUCCESS in 45m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/702a2786eb9b43d1a6b176a76e098b70 : FAILURE in 4m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/398ebb86397546c58b39702f6b890aef : SUCCESS in 1h 18m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a0fa8c3e532a4aa6901313db9ca98339 : SUCCESS in 1h 24m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d0f42aa878b34d45a0c68588b7f666f5 : SUCCESS in 1h 43m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c53df27a01a34fef8f85e9756369ffb4 : SUCCESS in 1h 06m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/595ab0953564491b942d514849ad5b14 : SUCCESS in 1h 37m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6c1e8395f8aa4b509efcd124407d2977 : SUCCESS in 47m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4309d0b562e14120b2419fe883b03eb6 : SUCCESS in 1h 18m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ac0c0901ad7942a7ae9bc8203a5b8ac6 : SUCCESS in 2h 23m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0fdf7cc9a5141ea9edcbaefa4f40437 : SUCCESS in 53m 29s","accounts_in_message":[],"_revision_number":6},{"id":"c1803e1de02542141d4760b7db4dbb798adc6423","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-06 13:05:29.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815660/6/iscsi/ : SUCCESS in 1h 11m 08s\n- cinder-smb http://cloudbase-ci.com/cinder/815660/6/smb/ : SUCCESS in 1h 14m 24s","accounts_in_message":[],"_revision_number":6},{"id":"4d502f88ca6ce255a698b969dc89aef32a11323c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-06 13:53:46.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/815660 : FAILURE in 1h 00m 58s","accounts_in_message":[],"_revision_number":6},{"id":"1ca88a26958344ab318d069cdcf9a40578184aea","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-06 14:07:28.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/815660 : FAILURE in 24m 26s","accounts_in_message":[],"_revision_number":6},{"id":"fa596702bb52a175afbadffb057a4bb1198142b0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-06 14:21:53.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/60/815660/6/check/iscsi-driver-alletra-6k/0d8bdf6/0d8bdf6.html : FAILURE in 1h 11m 07s","accounts_in_message":[],"_revision_number":6},{"id":"fd15e5811d328f138c2ca6f455ad5658eda107bf","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-04-06 14:26:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5be7a6d263fb92684e1cd83af29c9b7a91a37d28","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-04-06 14:31:42.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"eb07904c3030571f47ac500de0ef876e0f9ad6a1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-04-06 14:39:58.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d53c90ea0f8931440ca99877afd54f00e7e620d8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-04-06 14:41:01.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d67d1222d6830dd4253a3030fd941323efd865d9","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-04-06 15:04:05.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f9eb9f745a4db169559a5d666e70b299b0e7e4cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-04-06 15:28:00.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"935d18a5888b444bc2269a175468b8b32751054a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 15:52:32.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a825771d4d644c31bf7031f9e5b78f0f : SUCCESS in 18m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aaf85d49d4d747b2bffef073e3ad80a3 : SUCCESS in 22m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fa4ad4b2abec9118e1d6ca1924207fccc81162c5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-06 16:47:54.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815660/7/iscsi/ : SUCCESS in 1h 01m 42s\n- cinder-smb http://cloudbase-ci.com/cinder/815660/7/smb/ : SUCCESS in 1h 18m 24s","accounts_in_message":[],"_revision_number":7},{"id":"b35e31779d0a2c52d736d849cf17a2795a82cfd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 17:50:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/060991d7c0354060afa3166ad204f9ca : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99b216db88e545fe94c8114ad31f97e4 : SUCCESS in 9m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5299585fafb4786892a3ce5c90dbd62 : SUCCESS in 11m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/739eb72fa01847038893fda8f186c080 : RETRY_LIMIT in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84832256d16144e8a8888d82012ffc1a : SUCCESS in 11m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/42cf3da1a4f941e48219efb967282666 : SUCCESS in 50m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/828da3a2dcb74abfa705ee2302dc86a5 : SUCCESS in 1h 14m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ef7ffada7a4425c8f4b75fed06963b7 : SUCCESS in 41m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fe58dc253e3643b0ad41fd7b66347dda : SUCCESS in 13m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a5b6411f109343108f2708fc7e4b944a : SUCCESS in 5m 30s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2e3ea1986510453d83bdd9fe13e70a20 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/192644321c834a67afed2040bc357126 : SUCCESS in 6m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/764b9c9701944d18ac6eb33bd72a92d4 : SUCCESS in 7m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6c34a51f6af049659bde21e39151db67 : SUCCESS in 35m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7dcac46b50be465aa14f33967766ce01 : FAILURE in 4m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8d7bcfbe224944bc9d084d5e8b28c6ff : SUCCESS in 1h 07m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3808d55f7b15497a9f8a39afa3f2e9b1 : SUCCESS in 1h 16m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dff259f6fec342b2ac60d9c986bbd6b8 : SUCCESS in 1h 35m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/abac376b84a046d8a7e7f6b07992f202 : SUCCESS in 1h 02m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/97f84eeb93f741c2bd369c11acf27fb6 : SUCCESS in 1h 15m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/26e4d72b4c784233aca5b84cdf47daa6 : SUCCESS in 38m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/59934d1c177042f78e29428ff4b2f01d : SUCCESS in 1h 16m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f0c185b32c644bafb5ebab5534885196 : SUCCESS in 2h 06m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f27d12038304a5397620dd31cd910c0 : SUCCESS in 44m 38s","accounts_in_message":[],"_revision_number":7},{"id":"6adb5e13bf0afb7f89fc38328b5bf79df2f84469","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-04-06 17:51:50.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/7/upstream-solidfire-check/cinder-solidfire-iSCSI/eab9857/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"e424d416245591934f0b785e2ae0417324ec613a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-04-06 17:55:16.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/7/upstream-check/cinder-cDOT-iSCSI/fc38761/ : FAILURE in 2m 31s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/7/upstream-check/cinder-cDOT-NFS/55ada05/ : FAILURE in 3m 52s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/7/upstream-check/cinder-cDOT-FCP/d6b8055/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"140af1997f8e044ec4f4befc933481d4544f2351","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-06 18:00:21.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/815660 : FAILURE in 54m 06s","accounts_in_message":[],"_revision_number":7},{"id":"318f724ab7fc1939d765831d151250d4d9db39ec","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-04-06 18:07:43.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-60-815660-7/1649267416 : SUCCESS in 17m 25s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":7},{"id":"88e7ffd7d673dd618c826450e764a2c022f8324f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-04-06 18:22:56.000000000","message":"Patch Set 7:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/06_04_2022_17_50_24\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"338ec6ebb3a3d0cab793cd77557b99ef3dfaeca4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-06 18:49:20.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-60-815660-7/ns5_nfs : FAILURE in 25m 5s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-60-815660-7/ns5_iscsi : ABORTED in 1h 40m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"6701d17329c5f9cb58eda88e9ab51e41c6ed8c6e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-06 18:54:43.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-60-815660-7/ns5_nfs : ABORTED in 3h 24m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-60-815660-7/ns5_iscsi : ABORTED in 3h 24m 34s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"4328147da6b85e82a9fe172e006b267ee007520c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-06 18:56:16.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/60/815660/7/check/iscsi-driver-alletra-6k/693d202/693d202.html : FAILURE in 1h 01m 21s","accounts_in_message":[],"_revision_number":7},{"id":"9ef82414b304c48a18e8b8c1d06a5443ab1f77e1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-06 20:18:07.000000000","message":"Patch Set 7:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15736/","accounts_in_message":[],"_revision_number":7},{"id":"c32a7046023e8bc28eed6b97acf1402103af404a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-06 20:19:57.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15736/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"996f6e84acc314852d9334375624f0add5c71bfd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-06 20:20:27.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/d3ce162eab6f4f6d8b2fc62987e24ae8 : TIMED_OUT in 3h 08m 04s","accounts_in_message":[],"_revision_number":7},{"id":"c4c73b30f00f60a4ea2fd238dfd7edc8db8679c5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-06 20:28:28.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/815660 : SUCCESS in 1h 19m 18s","accounts_in_message":[],"_revision_number":7},{"id":"7a1f410a2a981a8c21ab5b18eeaee29bfebdf6ae","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-04-06 21:21:25.000000000","message":"Patch Set 7:\n\nBuild failed.","accounts_in_message":[],"_revision_number":7},{"id":"0dbe400861b24bc6f11952caa0a999798ce499ba","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-04-06 21:26:09.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/60/815660/7/check/DellEMC_PowerFlex/288b9ed/DellEMC_PowerFlex/538 : SUCCESS in 1h 33m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"370563b2a6658f132946e8a8f7c90f183442cb65","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-07 00:07:30.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/815660/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/aeac163/ : FAILURE in 2h 53m 14s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/815660/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b8d9768/ : SUCCESS in 1h 21m 16s","accounts_in_message":[],"_revision_number":7},{"id":"6415faf63108ccc1b47e7a6cfac0b8cd6c9a77e7","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-04-07 00:32:29.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/60/815660/7/check/DellEMC_SC_FC/8369fa7/DellEMC_SC_FC/406 : SUCCESS in 1h 03m 53s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/7/check/DellEMC_SC_iSCSI/b11f771/DellEMC_SC_iSCSI/427 : FAILURE in 2h 02m 35s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"328a4f530a6389ddc5db9f6d399c4cf72b0a5341","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-07 02:29:20.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-60-815660-7/ns5_nfs : SUCCESS in 3h 33m 40s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-60-815660-7/ns5_iscsi : SUCCESS in 3h 24m 34s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"53d026eca6bcd0da95df810d93e2df9d9b0e5fe6","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-04-07 03:49:48.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/815660-7/Lenovo-iSCSI/ : SUCCESS in 1:01:06\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/815660-7/Lenovo-FC/ : SUCCESS in 59:44.88","accounts_in_message":[],"_revision_number":7},{"id":"b707f23daa991d6343f05a4da3193f2b76bcc676","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-04-07 04:34:36.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/815660-7/PVME-iSCSI/ : SUCCESS in 1:07:10\n- PVME-FC https://os-logs.seagate.com/dell-ci/815660-7/PVME-FC/ : SUCCESS in 1:04:03","accounts_in_message":[],"_revision_number":7},{"id":"d4268cbf6ef62bb3c98df09e112672e12beac087","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-04-07 05:01:57.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/815660-7/Seagate-iSCSI/ : SUCCESS in 1:09:04\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/815660-7/Seagate-FC/ : SUCCESS in 1:04:27","accounts_in_message":[],"_revision_number":7},{"id":"cb6423185e7640ec278948ca83b0afba24ce0b1f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-04-07 05:37:22.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/14bc07de0c9c4d269cd8813a0cbb54f5 : FAILURE in 1h 02m 23s","accounts_in_message":[],"_revision_number":7},{"id":"c221ed36c6f5a958603a81844ab6950b1a84436e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-07 06:48:05.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/60/815660/7/check/dsvm-tempest-cisco-zonemanager-job/7824bb2/ : SUCCESS in 1h 13m 34s","accounts_in_message":[],"_revision_number":7},{"id":"0137e0b84baef3d30b089ee35de6f178e1f5c0f2","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-04-07 10:59:38.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ac034c0678917d664ed653d31668d0d2a52c3109","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-04-07 13:19:41.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/815660/7/ : SUCCESS in 2h 27m 26s","accounts_in_message":[],"_revision_number":7},{"id":"fafde8e517483b02d1dbf0fd45d10c7439e93b98","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-07 18:18:08.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/815660/7/check/iscsi-driver-nimble/83c909d/83c909d.html : FAILURE in 1h 01m 34s","accounts_in_message":[],"_revision_number":7},{"id":"97e2cd85df63e0e8cd18560ce67d54d1b2c47469","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-07 19:21:48.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/815660/7/check/iscsi-driver-nimble/06321f5/06321f5.html : FAILURE in 1h 01m 53s","accounts_in_message":[],"_revision_number":7},{"id":"58cdf3344f543226de2b1ce8f4613802fc4a053a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-04-07 19:51:59.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d324cfd506d823049100e477043f4c32521ba110","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-04-08 10:13:01.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"079bdc25930f3c159c50b9959f2fbd61e304a7d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-04-08 10:13:32.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"471ed0b16336d7e538e7697c3a5b3bb5964e9bfe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-08 10:13:40.000000000","message":"Patch Set 8:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15762/","accounts_in_message":[],"_revision_number":8},{"id":"ded0cfd8c75ce14b64b932f77a5e586aaaf67209","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-08 10:15:26.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/815660/8/check/DellEMC_Unity_FC/9e1891b/DellEMC_Unity_FC/534 : FAILURE in 1m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/8/check/DellEMC_Unity_iSCSI/a376cd5/DellEMC_Unity_iSCSI/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"92bc4ee37da4a4abb075ad8765d6c154140d734f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-08 10:15:31.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15762/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"bba585a63a954054366e7e575ab96a80a30cb8b3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-08 11:01:02.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/815660/815660/8/check-not-vote/ext-cinder-zuul/5b8981c : SUCCESS in 45m 32s","accounts_in_message":[],"_revision_number":8},{"id":"b38e8935fce8d7aaf7b730a08ef06e43234692d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 11:32:36.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0dde94d224e4403975e4fbe03cb53a2 : SUCCESS in 21m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8364ede372fb4b5ba6b56b60bd2116d6 : SUCCESS in 20m 56s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5b46703e04d701052ca5576aae5792fc1a2c78ae","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-08 12:16:03.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815660/8/iscsi/ : SUCCESS in 1h 08m 13s\n- cinder-smb http://cloudbase-ci.com/cinder/815660/8/smb/ : SUCCESS in 55m 10s","accounts_in_message":[],"_revision_number":8},{"id":"e18f23836a1b6147ffff0721ca2f8c34f442409b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-08 12:27:54.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/815660 : FAILURE in 57m 46s","accounts_in_message":[],"_revision_number":8},{"id":"1e73c19818e9fddc0e326dcafa9d2f7212e696be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 12:36:51.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/223798896352497280373543ecdc6694 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12ced45245ff4e64af2a2fc85f114ccc : SUCCESS in 10m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/858a5380292c422da1fc73302dce248b : SUCCESS in 12m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f0bfe3501544001bc80fab9cb90c398 : RETRY_LIMIT in 4m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcf64061e4b044e9a270485f64e437ab : SUCCESS in 12m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/8694dd4024bc4627a7bbe17b6bac655e : SUCCESS in 1h 05m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9a5e98706fab4cf59476b394e6ed0bc1 : SUCCESS in 1h 15m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73597289ffa145bca7febeb04d7fab32 : SUCCESS in 51m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b0a5c58016eb469c97261395e1ffa1f9 : SUCCESS in 16m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/51e614649da14da09e31904f64faef7a : SUCCESS in 6m 14s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9ed58fdc56bf444fa26425a87f9e6c25 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c0418e106cad487c88f0b641b8ca13d7 : SUCCESS in 6m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/02ef96073fd948f99b139ea56f16fc59 : SUCCESS in 7m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/04a6a4ab219a4ca2b7726482eebabe25 : SUCCESS in 38m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/69e80a397d31431c9ed87b5dec995496 : FAILURE in 4m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/34c3fca4a59148828146f00547b29e35 : SUCCESS in 1h 09m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6130cf9de32447a2a85344fb5628d8b5 : SUCCESS in 1h 22m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bccbd7e3b997400285cf11a8b68e4a49 : SUCCESS in 1h 50m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b8c2dc8b1b724d3a9350727ecc52177f : SUCCESS in 1h 06m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d5918afc00e44a13b87a4b6d031b81ff : SUCCESS in 1h 21m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4de51de7039245dc8bf00bcdb89e6532 : SUCCESS in 44m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5f20d1e2964448f6bc43bcd34b8f515b : SUCCESS in 1h 14m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1f951655c0bf48c78dd8cf957920c52d : SUCCESS in 2h 20m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b84fc6cba33c41309ac499fac9dee32e : SUCCESS in 52m 01s","accounts_in_message":[],"_revision_number":8},{"id":"1b689aef0baac10361bbcfbeab6cc664bddb7bfa","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-04-08 12:38:15.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/60/815660/8/check/DellEMC_PowerFlex/3b73262/DellEMC_PowerFlex/550 : FAILURE in 1m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"d3ec9c1d40c783925a5be5730d40a27ef2500206","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-04-08 12:38:29.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/8/upstream-solidfire-check/cinder-solidfire-iSCSI/6ba9b45/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"df19ad77d351002193f1dbba4a9681f86723fc17","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-08 12:38:45.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/815660/8/check/DellEMC_Unity_FC/7630a54/DellEMC_Unity_FC/537 : FAILURE in 1m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/8/check/DellEMC_Unity_iSCSI/bbca757/DellEMC_Unity_iSCSI/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"a23d4670f7928cdbecd0569e745b85744999171d","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-04-08 12:38:47.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/60/815660/8/check/DellEMC_SC_FC/9711fb7/DellEMC_SC_FC/418 : FAILURE in 1m 16s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/8/check/DellEMC_SC_iSCSI/69d42b3/DellEMC_SC_iSCSI/439 : FAILURE in 1m 35s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":8},{"id":"bcf3276f360c016c4c7a39c1555ab4a6f2c3b6c1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-04-08 12:43:09.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/8/upstream-check/cinder-cDOT-iSCSI/0d9ccf4/ : FAILURE in 2m 05s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/8/upstream-check/cinder-cDOT-NFS/bb50996/ : FAILURE in 2m 44s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/8/upstream-check/cinder-cDOT-FCP/6b0152d/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"32ed922464c32ea41ebc5ae4d6ba3981fa5c6f1c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-08 13:10:56.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/60/815660/8/check/iscsi-driver-alletra-6k/850c142/850c142.html : FAILURE in 1h 01m 41s","accounts_in_message":[],"_revision_number":8},{"id":"262cd280af444a9c56615a2e35f830ef611dcce6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-08 13:54:30.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/815660 : FAILURE in 1h 22m 35s","accounts_in_message":[],"_revision_number":8},{"id":"883bf2d64716f0dfaa27f125cbcf58c91b12ef9f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-08 14:02:45.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/ea3d1d60d4aa4cb085e322dd4673bca8 : POST_FAILURE in 1h 54m 05s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/cfc5e4a0ff314cb68db1efadb68d97f3 : TIMED_OUT in 3h 07m 35s","accounts_in_message":[],"_revision_number":8},{"id":"63f5ce7db58e796ccc5b5681e458f7b42f24b3b0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-04-08 14:04:28.000000000","message":"Patch Set 8:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/08_04_2022_13_22_34\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"581ea6eb502f3214cd46c8e5aa264e2e29781ff8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-08 15:02:29.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-60-815660-8/ns5_nfs : SUCCESS in 3h 25m 39s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-60-815660-8/ns5_iscsi : SUCCESS in 3h 28m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"55088f0764ccc71b0cab43f49abeb8403956070e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-08 15:31:19.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/815660/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/72ed694/ : FAILURE in 2h 22m 27s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/815660/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0431a19/ : SUCCESS in 1h 07m 25s","accounts_in_message":[],"_revision_number":8},{"id":"50c176df01f76da5ac97051f58c78a9e8650efc6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-08 17:01:43.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/60/815660/8/check/dsvm-tempest-cisco-zonemanager-job/afe7f27/ : SUCCESS in 1h 10m 50s","accounts_in_message":[],"_revision_number":8},{"id":"4c0175ee60383c88b19601e5dac66c56a9e45021","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-04-08 17:15:08.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/613420b2027b4063bbe10cc337d5788e : FAILURE in 1h 04m 55s","accounts_in_message":[],"_revision_number":8},{"id":"c2497aacbcb5d2efd5e9a91a39f797a1817c6322","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-04-08 18:48:23.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/815660-8/HPMSA-iSCSI/ : SUCCESS in 53:51.96\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/815660-8/HPMSA-FC/ : SUCCESS in 46:02.10","accounts_in_message":[],"_revision_number":8},{"id":"714b21bbeda9d5bc65fe812cc80bc2ea5fb38e55","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-04-08 20:01:15.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/815660-8/Lenovo-iSCSI/ : SUCCESS in 1:03:54\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/815660-8/Lenovo-FC/ : SUCCESS in 58:59.11","accounts_in_message":[],"_revision_number":8},{"id":"8f2482db76b694d4aaf6efa1a9843822512b1aa4","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-04-08 20:31:20.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/815660-8/PVME-iSCSI/ : SUCCESS in 1:05:44\n- PVME-FC https://os-logs.seagate.com/dell-ci/815660-8/PVME-FC/ : SUCCESS in 1:02:55","accounts_in_message":[],"_revision_number":8},{"id":"1ab549c3078ae077a9c5d7755bba96009f201aa2","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-04-08 20:39:18.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/815660-8/Seagate-iSCSI/ : SUCCESS in 1:06:53\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/815660-8/Seagate-FC/ : SUCCESS in 1:04:48","accounts_in_message":[],"_revision_number":8},{"id":"342951af482a2109642b8a04a295b21ce41411c3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-09 06:11:41.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/815660/8/check/iscsi-driver-nimble/56f72f5/56f72f5.html : FAILURE in 1h 01m 14s","accounts_in_message":[],"_revision_number":8},{"id":"dd5a71aa1f72385aaebbf45222fd403066da6327","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-09 07:15:00.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/815660/8/check/iscsi-driver-nimble/e24f183/e24f183.html : FAILURE in 1h 01m 31s","accounts_in_message":[],"_revision_number":8},{"id":"2604ed30c9ac2d77682d77d761d2e418fa21900a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-04-09 09:24:34.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/815660/8/ : SUCCESS in 2h 12m 48s","accounts_in_message":[],"_revision_number":8},{"id":"1e548de1452cae2b5b1815e0d290359e53d63961","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-04-10 07:50:08.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/8/20220410143712/NEC-V-FC : SUCCESS in 1h 49m 11s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/8/20220410143712/NEC-V-ISCSI : SUCCESS in 2h 12m 52s","accounts_in_message":[],"_revision_number":8},{"id":"32b1c137760bc0f898805997f949816602ebff31","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-04-11 12:08:55.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/60/815660/8/check/DellEMC_XtremIO_FC/bd7e210/DellEMC_XtremIO_FC/950 : FAILURE in 2m 14s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/8/check/DellEMC_XtremIO_iSCSI/82256c4/DellEMC_XtremIO_iSCSI/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"88b81b55e8d53f6ade5a3f381853dfaee931ef9b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-04-11 15:46:03.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/8/20220411212416/NEC-ISCSI : SUCCESS in 3h 19m 39s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/8/20220411212416/NEC-FC : SUCCESS in 3h 21m 44s","accounts_in_message":[],"_revision_number":8},{"id":"46d4eee161d25a27ee59a65e4218f17b33c802dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-05-12 18:03:06.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"55b4637f7af14e4b21bb2f768350633a253835bf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-12 18:07:02.000000000","message":"Patch Set 9:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"533d488ac3015601f26b164b0ee4cf52c8e723b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 18:27:31.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67af1bbc3c5a461e8e4336f64c9e37bc : SUCCESS in 19m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba3f664b6cd34da78a4de4c96505f2d7 : SUCCESS in 22m 08s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"97875ce50ce538152cebd7d42eea438d608586f6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-12 18:32:31.000000000","message":"Patch Set 9:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1652378594337 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"045a06ea1bb5d355a0bfc519e0372d99232b9038","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-12 19:24:29.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/815660 : SUCCESS in 1h 20m 01s","accounts_in_message":[],"_revision_number":9},{"id":"1029a9038c71df9358e1983db407f2eceb1665dc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-12 19:27:59.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815660/9/iscsi/ : SUCCESS in 1h 18m 36s\n- cinder-smb http://cloudbase-ci.com/cinder/815660/9/smb/ : SUCCESS in 1h 24m 14s","accounts_in_message":[],"_revision_number":9},{"id":"6856fcb5d8d8aada8cd452847501dea3f5c6a89b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 21:16:24.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/626eb54418f54ffba8e51789e37bb0c6 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b71d2f7ced4b4a6c829f0b43f93560d9 : SUCCESS in 9m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3388807476fb46bbb098a5dd7ef6d60f : SUCCESS in 12m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9267ee2e9a5d4d5fa5ee02520a04dbee : SUCCESS in 12m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0145e284a77f4b198e8b0cf08eb9f2e4 : SUCCESS in 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/195f863d8d1648a181778bfba0baa709 : SUCCESS in 1h 03m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/691edfd583a043d6aa8f8a44fff68291 : SUCCESS in 1h 07m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82c973ca9fcf4506b4fed160e4d4d5ae : SUCCESS in 52m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e93de03717c64ef79ab39c61abae30da : SUCCESS in 13m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/adf5aabf1ed44302b116c517627d4212 : SUCCESS in 5m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/654602e56ade469b8e9fc85b6cac1564 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a88577d63a8242d6bf9330b5adc53316 : SUCCESS in 7m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ad8a2a855fd445249bc09f0735594803 : SUCCESS in 6m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c80ccc71089b428389bc296224e73218 : SUCCESS in 45m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d7275bbb156242f5a242a1122038bac6 : FAILURE in 7m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/77fcccabc69c4b08b936ee4470466a12 : TIMED_OUT in 2h 04m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0342a08b977d46efa02bce3b4390e075 : FAILURE in 1h 28m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8f9d270269d34ef891e19f73baceb18d : SUCCESS in 2h 16m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e3eab6b241404f58a26add50298cdae8 : SUCCESS in 1h 13m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/72d548101b944017a2222ec5739dc1f7 : SUCCESS in 1h 26m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5ea114d5670544c5b92a66304dc144d9 : FAILURE in 44m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6f847a709ce148e98c0a0a111e62634d : SUCCESS in 1h 22m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a373189d29404c4b9fe414496a135b91 : SUCCESS in 3h 04m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/479cac2e2da74d9f962b032ec9a0881c : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":9},{"id":"e776558cf4cc176129862aa6e54d9071d330a622","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-12 21:51:15.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"83668c8fc29f6fd14d6df4fd7b28073cef100a8f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-12 22:06:03.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/2940dba225e543f2b11a7c69242821ff : POST_FAILURE in 1h 54m 19s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/abe3666ea5134b4baf28121127358a1e : TIMED_OUT in 3h 13m 30s","accounts_in_message":[],"_revision_number":9},{"id":"3ab45bf4f49efd58b6f70be870de75d0d1d0359a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-12 22:34:33.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/815660/9/check/iscsi-driver-nimble/16a55e8/16a55e8.html : SUCCESS in 1h 36m 03s","accounts_in_message":[],"_revision_number":9},{"id":"a337f6987f5bcf3a9ecc7e8471f7487d6792ca43","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-12 22:45:55.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/815660/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/7d22986/ : TIMED_OUT in 3h 03m 10s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/815660/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b4583a5/ : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":9},{"id":"8875e788cfd272d8b387a564c04bb5044e5203da","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-12 23:08:28.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-60-815660-9/ns5_nfs : SUCCESS in 4h 44m 55s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-60-815660-9/ns5_iscsi : SUCCESS in 4h 51m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"fee0593b0554cb802fa80123e25713b263ce2afa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 23:57:46.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/283f6afd178f41baa888c893e169227f : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e9b2b7ffae14e059bbe320ed702567e : SUCCESS in 13m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b68d68e33b6445fba50256eff52fc3d : SUCCESS in 13m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d0fd85700074238bdd655b9a3c4586b : SUCCESS in 13m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec3bf684bf2144cb99e4dd4df7abb665 : SUCCESS in 13m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/0171c988dec14d649d9240dea4d4a2e2 : SUCCESS in 58m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6bfa4afab88841d689d023160088c8e6 : SUCCESS in 1h 22m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27c0def69c03454f86162b8ccbebe448 : SUCCESS in 44m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ed263ed5e2214f7589cd7ff7d7400afa : SUCCESS in 16m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7cf24c5fcf2c48f198d9536cc5d92d19 : SUCCESS in 5m 36s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6cb70bda26ff476eb2f75e8bf1c777e8 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a1f58182cd42440b9177a1df7317174b : SUCCESS in 6m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f875c3abe28d4d22a8dd8cdd150f33b3 : SUCCESS in 7m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/50d8848e22ec44e086395a9c4369a5f4 : SUCCESS in 42m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b99b23cad67e45a4a2f68d620227525a : FAILURE in 7m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9e7c9b161943458a838dfa1626f77393 : FAILURE in 1h 14m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d72e9f4f6f2b49a18653207c02da5262 : FAILURE in 1h 24m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3daaf228e97e4745a869af5177164188 : SUCCESS in 1h 32m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1c28e592bc084541ad967e4623d8314c : SUCCESS in 1h 01m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e2a3b8767aee4feb8f0563d4043d7080 : SUCCESS in 1h 11m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ec5fddfe8ddf4f2d8d591cc2e21dd4bf : SUCCESS in 56m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3ce07c685d774348bab2a22eb6aca964 : SUCCESS in 1h 20m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c46ae73995b84685ac9c3873d86c90e0 : SUCCESS in 2h 05m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4d920ac85de4bd098b974d74d909f4e : SUCCESS in 47m 07s","accounts_in_message":[],"_revision_number":9},{"id":"426dd7cd133a646d9338a5bd0dc4de0cae5b3c78","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-13 03:23:12.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/369436e96a17475bbe2e4009a27a462c : SUCCESS in 1h 29m 07s","accounts_in_message":[],"_revision_number":9},{"id":"001fb6ded8a7fe210b5a061b45197180d72fcd80","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-13 13:33:17.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"bdde14facf880f732a2b79db5fd1250591417fd3","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-05-13 13:34:37.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"679c4ef597177fcdcc168f7808303a26cf2cf8f5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-13 13:39:11.000000000","message":"Patch Set 9:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"3f50cf1ff805158f7d7908a97431b0532e5d44f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 14:00:39.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5712f7aa3cd5465ead37cb7c5989e291 : SUCCESS in 23m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/52ababd1aaef4d11873997322d27698a : SUCCESS in 21m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ccbe3a5d981ad6d9dadf8643362ffb54cedd80c9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-13 14:47:07.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815660/9/iscsi/ : SUCCESS in 1h 08m 15s\n- cinder-smb http://cloudbase-ci.com/cinder/815660/9/smb/ : SUCCESS in 56m 18s","accounts_in_message":[],"_revision_number":9},{"id":"f460e734cc99cba06e82bc723eaaf7b22233ddbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 16:09:50.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99ce96148d7248f0a57c8b58f76b4010 : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4696faf5817d412aabb05b297a81f876 : SUCCESS in 12m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9ce907fc88f450bbb6d14e0db2884bc : SUCCESS in 12m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/301788e8344f46a086169d505f8398c2 : SUCCESS in 11m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f6a7d92e25a4be08966ec9ff4deca29 : SUCCESS in 12m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/4d4d5d41e74b48bf86801964037f4e6d : SUCCESS in 52m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f77914eab88740c9bd8282271afa1a55 : SUCCESS in 1h 19m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e98181e5d8c34181b4b052b6fc867d8c : SUCCESS in 46m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/30046bfeef2c4a66ac733faca3ad8240 : SUCCESS in 14m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/95ccf151908441e1a8670f43d0110eca : SUCCESS in 6m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/224dec08928b4dbebd677f753996dfef : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/664f053a3d1e48b29fad531b4a9b2f1f : SUCCESS in 6m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6dc6f7bc24bd4fc58958e8f0304cfacb : SUCCESS in 8m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d2d1a8c28f494bc4bc520e6e8df4374e : SUCCESS in 57m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9478ba146c0443acb3e98775a31222ae : FAILURE in 7m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ae5080bf5cdb4c658de1d7d73746124b : FAILURE in 1h 31m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/086f30927eb246a3b096d3b902154e17 : FAILURE in 1h 36m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7075fcfe00fb4625a5f2dd539add04af : SUCCESS in 1h 48m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c4e243d653584dc6b51ba7f5b632befd : SUCCESS in 1h 24m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ba2bd86e627d47dd8cc3ce5123287099 : SUCCESS in 1h 48m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dede4c59d8754783b21e03bb2de7d942 : SUCCESS in 37m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/67b93d49606d45b9bb7a106037434eba : SUCCESS in 1h 34m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4d67271b47d2409e9510b026135f4a99 : SUCCESS in 2h 30m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f4cc86b7b374715a6aae471c8d2cab8 : SUCCESS in 57m 07s","accounts_in_message":[],"_revision_number":9},{"id":"ebd827c75a054576aa171e23e67926dfe2c6344c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-13 16:40:26.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/ea76ce8cdf534c2eaafabf405ac02373 : POST_FAILURE in 2h 09m 14s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/e07b951377754614aac01df9e702845a : POST_FAILURE in 2h 38m 09s","accounts_in_message":[],"_revision_number":9},{"id":"9b51a9b766a6dcab6c0df52ffffb9fda9a1a663f","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-05-13 18:01:01.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/c39b3b4/index.html : SUCCESS in 1h 30m 25s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/55961f1/index.html : FAILURE in 1h 37m 59s","accounts_in_message":[],"_revision_number":9},{"id":"fb1361a51f597f9241006d050ff272d3f58d8eca","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-13 19:01:55.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/9/20220514015350/NEC-V-FC : SUCCESS in 2h 00m 23s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/9/20220514015350/NEC-V-ISCSI : SUCCESS in 2h 08m 01s","accounts_in_message":[],"_revision_number":9},{"id":"380105740845279e92578077afdb6661cf6b0503","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-13 20:50:41.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/843447530734416194b4c4f73e877751 : SUCCESS in 1h 27m 48s","accounts_in_message":[],"_revision_number":9},{"id":"984da984a3205a9215f567566a42ef8c32ac6ff7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-14 01:26:44.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"dee2f088615b782b4f80677a0425c9c4fb0c3450","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-14 01:28:35.000000000","message":"Patch Set 9:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"d54eac26da373661ded7ed1e42a92df1106caa9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-14 03:38:23.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7716f092c0f94735885d5520bd4dd330 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b1442f144334c928686ba38072db009 : SUCCESS in 10m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1c8cae4c749413cb873067ec3fbb915 : SUCCESS in 11m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a628e4b3354490491cab26133b48638 : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14f97e3bc56d4ddb843d66905089fe08 : SUCCESS in 10m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/044c91ce0bbe49ceb7d572ed77f0199a : SUCCESS in 46m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/caee7d1cd0ad4f8781f0398df181072d : SUCCESS in 1h 06m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45831a54f17a40c286baea2cd03b5979 : SUCCESS in 45m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c00911957a5a47f9b88e9160c84088cd : SUCCESS in 13m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ff80be2bc95c461ba4383b568eb68d07 : SUCCESS in 5m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cf97f91ae2094884a9404486bc551001 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c9a37230a1c24dc8886ade638a032b00 : SUCCESS in 6m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/45083177c50141cb82316257470911be : SUCCESS in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7e11f75934ac4e23b6abb9c1bd9f9b67 : SUCCESS in 34m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a2878c0504f54236b1e3da6c552c9cb2 : FAILURE in 7m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f07db430b9124514ba035af58c909302 : FAILURE in 1h 24m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/127b342fd1204ed5a3a29f987b53a1d2 : FAILURE in 1h 10m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ae26da680e664d9b8caaee103f5496a6 : SUCCESS in 1h 50m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/29bbfabd53394140b53c27c3b81edc85 : SUCCESS in 1h 14m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1df2b77a06cd47b7a90d0a32d1c5b8ec : SUCCESS in 1h 27m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c5f30ac7c4aa424ab13a07ae0ace5665 : SUCCESS in 38m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0d326a38efd949b08003d4fc191b8978 : SUCCESS in 1h 27m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/45acd0c4d5934835a6b8b5188a98b88a : SUCCESS in 2h 08m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60a681f911b24a00ac40758c93740346 : SUCCESS in 39m 55s","accounts_in_message":[],"_revision_number":9},{"id":"3439c882dbf158a39f9f423a37697387a87a2300","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-05-15 11:08:19.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/9/20220515164929/NEC-FC : SUCCESS in 3h 17m 49s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/815660/9/20220515164929/NEC-ISCSI : SUCCESS in 3h 18m 46s","accounts_in_message":[],"_revision_number":9},{"id":"4924a6389fa919a5881326de5a34b69b5693355e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-08-10 10:45:29.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"44b475d25c15701391ff9ffa013942f1c8e33272","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 11:08:52.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b0281270cf24980ba5e8813705e05bf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/059459ffee8643cc8f02f9baae14529d : SUCCESS in 21m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b66b8bb15aac4ba7ab6568ad1a5ace2f : SUCCESS in 17m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b7dc94f02196e7e5bb96c5142def1e2f006e1a30","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-10 12:27:51.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815660/9/iscsi/ : SUCCESS in 57m 01s\n- cinder-smb http://cloudbase-ci.com/cinder/815660/9/smb/ : SUCCESS in 1h 00m 07s","accounts_in_message":[],"_revision_number":9},{"id":"ac7505c4cab78bd061b225f27299d869dcedce70","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-08-10 12:49:24.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi https://test-cinder99.s3.ap-south-1.amazonaws.com/storwize-tempest-dsvm-full-iscsi/aa1e3d3/index.html : SUCCESS in 1h 39m 42s\n- storwize-tempest-dsvm-full-fc https://test-cinder99.s3.ap-south-1.amazonaws.com/storwize-tempest-dsvm-full-fc/10dc095/index.html : TIMED_OUT in 2h 03m 11s","accounts_in_message":[],"_revision_number":9},{"id":"8720c980038cba4581774428fa4df40e5ff5a554","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 12:58:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c005cad3bf6489ea0106ac09770a9e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76ff9fac71514825927b655d19c3f1ca : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aebd169fc0174e19a99f82d9e94ce832 : SUCCESS in 11m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6991b012ada9434ca9d221a87724d2dd : SUCCESS in 13m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3db28eef13434193b2f0144d853ba6ec : SUCCESS in 11m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6af8eef095874b239170d0221ff66970 : SUCCESS in 11m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/a525e7cb584940ab97a34505a9511c82 : SUCCESS in 50m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/703641aaa3704abdb2bf9dc20fddfe08 : SUCCESS in 1h 04m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b875f3c816e4a97a4ea8733fdb4dac8 : SUCCESS in 50m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9c13892cd399472badbe81d82bcff733 : SUCCESS in 16m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/30454248f9b84f0a812e218dfc9a5f00 : SUCCESS in 5m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/90a255f0776b41118c3aedca1663cd83 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eca0db9f1c8e4dd48b247ccb592ec389 : SUCCESS in 6m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c719091858724e79ba0ca0d667ff1d4a : SUCCESS in 7m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e1c0ef26c5d8464db054b46a2433ba49 : SUCCESS in 34m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a26f0efad3864293a10563a8dd4ab78f : FAILURE in 6m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5c246c4428bc4fd7803a870cfaa30509 : SUCCESS in 1h 08m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1d6f623932624445baff9cf287a3383d : FAILURE in 1h 13m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/82435880a5fc4edf9816aa856318a54b : SUCCESS in 1h 19m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b19797a9d2084541a582de2ee6d3e623 : SUCCESS in 56m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/36271bc5d7694f69ac7fd552803520bc : SUCCESS in 1h 13m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f7312541eea64e17bbc3e43a8d2587b6 : SUCCESS in 34m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d6a81eb2a40d4cafa457a4555578e9f1 : SUCCESS in 1h 30m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3f5bed830b254b51a68481c770fe7fbf : SUCCESS in 2h 11m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf95686523a9455c8674aa7a139076a1 : SUCCESS in 55m 53s","accounts_in_message":[],"_revision_number":9},{"id":"8e49dbbad6a9ab2c23ce97318bf9e170982f9eb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 12:59:09.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"a6f76fe2eef09e00b755637d5bc697a04e377a79","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-10 13:46:14.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/60/815660/9/check/dsvm-tempest-cisco-zonemanager-job/7165823/ : FAILURE in 35m 52s","accounts_in_message":[],"_revision_number":9},{"id":"e98e1264d960641461e38fb37ca3e50902587695","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-10 13:57:08.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/9/upstream-check/cinder-tempest-plugin-ontap-fcp/cf8159d/ : RETRY_LIMIT in 9m 37s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/9/upstream-check/cinder-tempest-plugin-ontap-iscsi/abb9e5c/ : SUCCESS in 41m 50s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/9/upstream-check/cinder-tempest-plugin-ontap-nfs/dd7776f/ : SUCCESS in 44m 31s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/9/upstream-check/cinder-tempest-plugin-solidfire-iscsi/1ba2983/ : SUCCESS in 57m 38s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"770b04354c38abc96e2b619afc83e883ad2cdd12","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-10 14:19:04.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/815660/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/75a703f/ : SUCCESS in 1h 45m 07s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/815660/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ae38796/ : SUCCESS in 1h 40m 33s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/60/815660/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/fb13b33/ : SUCCESS in 1h 39m 11s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5ed018205778f5cb3d407fd3c2957c58d5e77444","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-10 14:43:58.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/55d245e5f9fa4786b60da70607ea79ba : POST_FAILURE in 2h 05m 50s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/8f4178c8e78e4ec28179115049f7e169 : POST_FAILURE in 2h 42m 54s","accounts_in_message":[],"_revision_number":9},{"id":"c7f367d44e87cb399d0f13b4e19d804c25c1306b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 14:48:53.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/464ac0a3e20f44f4922ebcc11c750684\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fa489a681b141cfbcc38ec9a66b84c3 : SUCCESS in 8m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ee4b2e527d54b2cbf8c7f70777493ef : SUCCESS in 12m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/442dada1eb8f49e5b1e9db695461d401 : FAILURE in 13m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa8c6bc2132411a936e2aa4726f5acb : SUCCESS in 13m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/9bb786bd5d664ee7a5994bb625e5203d : SUCCESS in 57m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5cfb77704c6f46aba05daf54cd126586 : SUCCESS in 1h 23m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72950ac48b2b41aeb6a6e9cef1f8a5b0 : SUCCESS in 52m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8532510af89242e5b075f9c6fa459526 : SUCCESS in 1h 12m 47s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/427aba3c276242b6991b0b0c8287ea2c : SUCCESS in 1h 19m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47a8da4c91154a5fb75b5f53ffa7badc : SUCCESS in 51m 07s","accounts_in_message":[],"_revision_number":9},{"id":"df0ba46f165ab9a61e693e31be8b8839614d3527","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-10 15:00:24.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/60/815660/9/check/DellEMC_PowerStore_FC/49b6e10/DellEMC_PowerStore_FC/715 : SUCCESS in 1h 21m 40s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/60/815660/9/check/DellEMC_PowerStore_NFS/752ee3d/DellEMC_PowerStore_NFS/256 : FAILURE in 1h 24m 24s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/60/815660/9/check/DellEMC_PowerStore_NVMeTCP/c0e6fe2/DellEMC_PowerStore_NVMeTCP/130 : SUCCESS in 1h 44m 25s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"6398696d89a5405eb47da0f28f2e2132052eb76c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-08-10 15:59:29.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/60/815660/9/check/DellEMC_SC_FC/10a824c/DellEMC_SC_FC/902 : SUCCESS in 50m 14s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/9/check/DellEMC_SC_iSCSI/918ee76/DellEMC_SC_iSCSI/920 : SUCCESS in 2h 13m 37s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":9},{"id":"b11082f56e0c53e8aa419e4f3882b669ae766056","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-08-10 16:05:35.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/815660-9/HPMSA-iSCSI/ : SUCCESS in 53:20.27\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/815660-9/HPMSA-FC/ : SUCCESS in 49:28.28","accounts_in_message":[],"_revision_number":9},{"id":"17bc27275c32a0f00954cfe49237b3dd36d42b15","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-10 16:07:15.000000000","message":"Patch Set 9:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/10_08_2022_13_50_01\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"dae329cceb95a0c11f9263be7be9ec5523550994","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-08-10 17:11:10.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/815660-9/PVME-iSCSI/ : SUCCESS in 1:09:16\n- PVME-FC https://os-logs.seagate.com/dell-ci/815660-9/PVME-FC/ : SUCCESS in 1:09:23","accounts_in_message":[],"_revision_number":9},{"id":"52ed14ab7b61c0ef91af65e83fc3e624922581b9","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-08-10 20:32:16.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/815660-9/Seagate-iSCSI/ : FAILURE in 1:20:44\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/815660-9/Seagate-FC/ : SUCCESS in 1:17:30","accounts_in_message":[],"_revision_number":9},{"id":"ecb7b6b111e11a957679bfc865ed616e2749eb4f","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-08-10 20:43:20.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/815660-9/Lenovo-iSCSI/ : SUCCESS in 1:07:51\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/815660-9/Lenovo-FC/ : SUCCESS in 1:04:45","accounts_in_message":[],"_revision_number":9},{"id":"7b8ef3f9638412b05233c503d1d598e84093b644","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-11 02:10:11.000000000","message":"Patch Set 9:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/422","accounts_in_message":[],"_revision_number":9},{"id":"0cef1f0dec8625987bc871b86a215aa6afb80b66","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-11 09:05:45.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"681c4c3c3a9ca216b74caaf90af385431f7113a6","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-08-11 09:20:17.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"14b9e87046c77368477716e0530e97e741834f85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 09:26:13.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aacd70812ff24706803a76637d2a4c47\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb09dbfb67af412ab254c43834d9c84d : SUCCESS in 15m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/343a1bfb629c4a3fb889505793ed4448 : SUCCESS in 15m 44s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8c1362b5c50b913f5d6a888671b6133ab8f00e6a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-11 09:50:52.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/60/815660/9/check/dsvm-tempest-cisco-zonemanager-job/f6391c9/ : FAILURE in 36m 10s","accounts_in_message":[],"_revision_number":9},{"id":"348ac8fdbee4c23068bba8a439cfa7494292d899","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-11 10:14:26.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/815660/9/iscsi/ : SUCCESS in 1h 08m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/815660/9/smb/ : SUCCESS in 1h 03m 02s","accounts_in_message":[],"_revision_number":9},{"id":"09551839cf0207b9b5ab3b4da74328195d7f4183","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 11:49:20.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6387cec421b46fe8081381372cc6e7f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdb1ad443c6d4b40b33c70103edd077d : SUCCESS in 10m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60f2992374584ccebeb938f6ed07ca73 : SUCCESS in 12m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b5f26235a3745c2a482c48569486e1d : SUCCESS in 13m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca1da99785b14c24a01857bd91f9d0c4 : SUCCESS in 14m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3196a206dfe24480a143d57cbf212dfd : SUCCESS in 14m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/029718dc983d40c8b25cf8c90d130a45 : SUCCESS in 1h 07m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/234d4cd2bfb84c85bc59039fc5ec3487 : SUCCESS in 1h 41m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae6d50f164174bb0a8621d9a79087d3b : SUCCESS in 1h 02m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7dabe3c769b1457089ffc7566eb5cf8b : SUCCESS in 16m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/13facbc718f442dca53a7dbd906a1bb1 : SUCCESS in 7m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d6ba5fa0bb6949e2a04464fe583b5702 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1828cdc32c1d43cb96af4c01d299e715 : SUCCESS in 8m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8418f709efe54ffcb9789df458bd699e : SUCCESS in 9m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8ff360c1cd514b7a88fe9a047190ed0e : SUCCESS in 49m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7eaf95c9986a4547a811eab0b9b67b76 : FAILURE in 8m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2257301030f54373b11a6c847388598f : SUCCESS in 1h 38m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e1a3cbf8b1d14e71aa59a05da6745950 : FAILURE in 1h 46m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/62dcbd63333b469c8ed58de586d50a3c : SUCCESS in 1h 39m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d9284923a6e64864a1f06898cd4e5f40 : SUCCESS in 1h 18m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2ae268c91ce14a5abb9c2cc6d2fcb35b : SUCCESS in 1h 37m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/54e130d7d59946d8a5194fa9ddd80680 : SUCCESS in 59m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d799ec80d1f3422eb8db7971b65919cd : SUCCESS in 1h 37m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3640677e32b143a89eac4e990632cf35 : SUCCESS in 2h 21m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4d4db2d4f12470cb67b3321cf108a0f : SUCCESS in 51m 23s","accounts_in_message":[],"_revision_number":9},{"id":"225ff22f56bef645193139fc67eaf36bf86234fa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-11 11:49:36.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/815660/9/check/tempest-dsvm-full-focal-py3/28dcf1f/ : FAILURE in 1h 59m 13s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/815660/9/check/devstack-plugin-ceph-dsvm-focal-py3/74413bd/ : FAILURE in 2h 40m 45s","accounts_in_message":[],"_revision_number":9},{"id":"8d9c16ff5fd122c67d91358c613286b2584283f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 11:49:59.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"5b16ff435bb91aacb6f544b0f0ba383715ee2c5c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-11 12:56:04.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/9/upstream-check/cinder-tempest-plugin-ontap-fcp/7c21029/ : SUCCESS in 1h 01m 14s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/9/upstream-check/cinder-tempest-plugin-ontap-iscsi/2887041/ : SUCCESS in 43m 36s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/9/upstream-check/cinder-tempest-plugin-ontap-nfs/29ec565/ : SUCCESS in 44m 11s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/815660/9/upstream-check/cinder-tempest-plugin-solidfire-iscsi/8b3b604/ : SUCCESS in 53m 08s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"9def279c0159540ee43de66bbf894be08118d58c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 13:23:49.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c4a10316b694734adfaf798896753ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/263e35799ed34b998f15c3e4c2ffb4b3 : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3af8b5273a8643d2a4e0761c7691e914 : SUCCESS in 11m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1dc98ce88f741198373ba4cb396683c : SUCCESS in 13m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca48f10259a34cc585847c7e119a4321 : SUCCESS in 11m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/12fad3b6944948ebb0879326bcf022be : SUCCESS in 1h 07m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d8cdcd89f7d0422eaa08dfa864a6ac14 : SUCCESS in 1h 32m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ccba159a3574d8ca60df1b13b28558c : SUCCESS in 54m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e1dd9f30314749fbb83e375719e4a5bf : SUCCESS in 1h 17m 35s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f3e131ad68134b03990be57491f03842 : SUCCESS in 1h 16m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc4318fbf2694f00afeae54059df8cb4 : SUCCESS in 58m 47s","accounts_in_message":[],"_revision_number":9},{"id":"5e15b61e9afac840d5e1a1aeaeee7fa95d2f9162","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 13:23:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"0b7f4556b18933d756932cfe7b9edf62d384ebe1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 13:25:04.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49f934eec2654089a8caf1a24e16cc95\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6325e34286e4416a90e57d5d3ad7a02f : SUCCESS in 54s","accounts_in_message":[],"_revision_number":9},{"id":"0f61d54f0e295e5c61e3047a8d0c4b6c4a482284","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-11 13:32:12.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/60/815660/9/check/DellEMC_PowerStore_FC/7fdf11d/DellEMC_PowerStore_FC/723 : SUCCESS in 1h 20m 09s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/60/815660/9/check/DellEMC_PowerStore_NFS/1b5b9f1/DellEMC_PowerStore_NFS/264 : SUCCESS in 1h 30m 02s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/60/815660/9/check/DellEMC_PowerStore_NVMeTCP/ded9656/DellEMC_PowerStore_NVMeTCP/138 : SUCCESS in 1h 42m 31s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"73b41c1902f9c9d73ae013f9d5d1952301fd2e78","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-08-11 13:45:53.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/60/815660/9/check/DellEMC_SC_FC/403ebb6/DellEMC_SC_FC/910 : SUCCESS in 52m 27s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/9/check/DellEMC_SC_iSCSI/536c5af/DellEMC_SC_iSCSI/928 : SUCCESS in 1h 56m 15s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":9},{"id":"6d553f30ed848215ecbc926d7983762be0653cbd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-11 13:50:16.000000000","message":"Patch Set 9:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/11_08_2022_11_49_29\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"0d537e0aaae941c51252ee5c4bb1b325d5b8284c","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-08-11 14:09:31.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/815660-9/HPMSA-iSCSI/ : SUCCESS in 55:31.79\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/815660-9/HPMSA-FC/ : SUCCESS in 53:17.50","accounts_in_message":[],"_revision_number":9},{"id":"acda17069a94b510b7504e347bfe8545c7f860b9","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-08-11 14:31:52.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/815660-9/PVME-iSCSI/ : SUCCESS in 1:05:57\n- PVME-FC https://os-logs.seagate.com/dell-ci/815660-9/PVME-FC/ : SUCCESS in 1:06:27","accounts_in_message":[],"_revision_number":9},{"id":"e6775acc8d08d4bb520e5e65590d7444d069f381","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-11 14:42:36.000000000","message":"Patch Set 9:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/436 \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":9},{"id":"e2403fb7dc657482ecfc40ee41133c7c8019868c","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-08-11 22:25:58.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/815660-9/Seagate-iSCSI/ : SUCCESS in 1:17:39\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/815660-9/Seagate-FC/ : SUCCESS in 1:20:21","accounts_in_message":[],"_revision_number":9},{"id":"91babe43d508284dd3a23fcdfba18353197b6262","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-08-11 22:28:24.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/815660-9/Lenovo-iSCSI/ : SUCCESS in 1:06:39\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/815660-9/Lenovo-FC/ : SUCCESS in 1:02:58","accounts_in_message":[],"_revision_number":9},{"id":"8259da3f6559708ea6a83d2d2fc9e190d9c72885","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-13 10:30:31.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/60/815660/9/check/DellEMC_Unity_FC/a824723/DellEMC_Unity_FC/1436 : FAILURE in 19m 12s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/60/815660/9/check/DellEMC_Unity_iSCSI/a479536/DellEMC_Unity_iSCSI/1336 : FAILURE in 1h 48m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"e071ee263201d5ed0689e0755cf06f0a8568d16c","revisions":{"68c89c7181b3b749402b5339c152cb4505408aab":{"kind":"REWORK","_number":1,"created":"2021-10-27 13:14:15.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/60/815660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/815660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/815660/1"}}},"commit":{"parents":[{"commit":"bbb315df34e274e5e17cefad3488085834753ea3","subject":"Join do_setup and check_for_setup_error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bbb315df34e274e5e17cefad3488085834753ea3"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-10-27 13:09:00.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-10-27 13:09:00.000000000","tz":120},"subject":"Docs: Document clone_image driver method","message":"Docs: Document clone_image driver method\n\nThis patch improves and fixes the documentation of the clone_image\ndriver method and updates the parameters in various places where the\nmethod is declared where they presented the wrong parameters.\n\nSince there is a valid default implementation for the method, a small\ntext has been added to indicate that some of the core methods, like this\none, are optional.\n\nTo avoid getting docstrings from the driver and the interface (which is\nthe one used for the documentation) out of sync we replace the\ndescription on cinder/volume/driver.py with a text referencing the\ninterface docstring.\n\nChange-Id: I09a8357e15b9041790285ab1f9a3cdd7a18243b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68c89c7181b3b749402b5339c152cb4505408aab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68c89c7181b3b749402b5339c152cb4505408aab"}]},"branch":"refs/heads/master"},"04fe235e556fca1d19731eb76a3efb3419859804":{"kind":"REWORK","_number":2,"created":"2021-10-27 13:20:20.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/60/815660/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/815660/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/815660/2"}}},"commit":{"parents":[{"commit":"bbb315df34e274e5e17cefad3488085834753ea3","subject":"Join do_setup and check_for_setup_error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bbb315df34e274e5e17cefad3488085834753ea3"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-10-27 13:09:00.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-10-27 13:20:11.000000000","tz":120},"subject":"Docs: Document clone_image driver method","message":"Docs: Document clone_image driver method\n\nThis patch improves and fixes the documentation of the clone_image\ndriver method and updates the parameters in various places where the\nmethod is declared where they presented the wrong parameters.\n\nSince there is a valid default implementation for the method, a small\ntext has been added to indicate that some of the core methods, like this\none, are optional.\n\nTo avoid getting docstrings from the driver and the interface (which is\nthe one used for the documentation) out of sync we replace the\ndescription on cinder/volume/driver.py with a text referencing the\ninterface docstring.\n\nChange-Id: I09a8357e15b9041790285ab1f9a3cdd7a18243b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04fe235e556fca1d19731eb76a3efb3419859804"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04fe235e556fca1d19731eb76a3efb3419859804"}]},"branch":"refs/heads/master"},"0f3fcbc6e19c88e6c714c3239a596c79a15c4152":{"kind":"NO_CHANGE","_number":3,"created":"2021-10-28 07:53:49.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/60/815660/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/815660/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/815660/3"}}},"commit":{"parents":[{"commit":"19b115e7d6f8fb8ff17e18249b7041827add8123","subject":"Join do_setup and check_for_setup_error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/19b115e7d6f8fb8ff17e18249b7041827add8123"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-10-27 13:09:00.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-10-28 07:51:03.000000000","tz":120},"subject":"Docs: Document clone_image driver method","message":"Docs: Document clone_image driver method\n\nThis patch improves and fixes the documentation of the clone_image\ndriver method and updates the parameters in various places where the\nmethod is declared where they presented the wrong parameters.\n\nSince there is a valid default implementation for the method, a small\ntext has been added to indicate that some of the core methods, like this\none, are optional.\n\nTo avoid getting docstrings from the driver and the interface (which is\nthe one used for the documentation) out of sync we replace the\ndescription on cinder/volume/driver.py with a text referencing the\ninterface docstring.\n\nChange-Id: I09a8357e15b9041790285ab1f9a3cdd7a18243b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0f3fcbc6e19c88e6c714c3239a596c79a15c4152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0f3fcbc6e19c88e6c714c3239a596c79a15c4152"}]},"branch":"refs/heads/master"},"4ee6e92fe227131f844133cd8859f72cca649d67":{"kind":"REWORK","_number":4,"created":"2021-12-02 18:33:30.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/60/815660/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/815660/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/815660/4"}}},"commit":{"parents":[{"commit":"f5956ecd331eb8e0a6d2ebc0e020cc813fb2c7e2","subject":"Join do_setup and check_for_setup_error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f5956ecd331eb8e0a6d2ebc0e020cc813fb2c7e2"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-10-27 13:09:00.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-12-02 18:33:21.000000000","tz":60},"subject":"Docs: Document clone_image driver method","message":"Docs: Document clone_image driver method\n\nThis patch improves and fixes the documentation of the clone_image\ndriver method and updates the parameters in various places where the\nmethod is declared where they presented the wrong parameters.\n\nSince there is a valid default implementation for the method, a small\ntext has been added to indicate that some of the core methods, like this\none, are optional.\n\nTo avoid getting docstrings from the driver and the interface (which is\nthe one used for the documentation) out of sync we replace the\ndescription on cinder/volume/driver.py with a text referencing the\ninterface docstring.\n\nChange-Id: I09a8357e15b9041790285ab1f9a3cdd7a18243b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ee6e92fe227131f844133cd8859f72cca649d67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ee6e92fe227131f844133cd8859f72cca649d67"}]},"branch":"refs/heads/master"},"e946909278113b76a5a54fe08d6c59f63423b070":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-01-31 15:59:44.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/60/815660/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/815660/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/815660/5"}}},"commit":{"parents":[{"commit":"a58bf7c0757a2de9470f3bb7a98898c5b3a3c4bd","subject":"Join do_setup and check_for_setup_error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a58bf7c0757a2de9470f3bb7a98898c5b3a3c4bd"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-10-27 13:09:00.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-01-31 15:42:17.000000000","tz":60},"subject":"Docs: Document clone_image driver method","message":"Docs: Document clone_image driver method\n\nThis patch improves and fixes the documentation of the clone_image\ndriver method and updates the parameters in various places where the\nmethod is declared where they presented the wrong parameters.\n\nSince there is a valid default implementation for the method, a small\ntext has been added to indicate that some of the core methods, like this\none, are optional.\n\nTo avoid getting docstrings from the driver and the interface (which is\nthe one used for the documentation) out of sync we replace the\ndescription on cinder/volume/driver.py with a text referencing the\ninterface docstring.\n\nChange-Id: I09a8357e15b9041790285ab1f9a3cdd7a18243b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e946909278113b76a5a54fe08d6c59f63423b070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e946909278113b76a5a54fe08d6c59f63423b070"}]},"branch":"refs/heads/master"},"08b75f26e6484c4461ea1e1be0cf081687665e25":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-04-06 10:11:15.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/60/815660/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/815660/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/815660/6"}}},"commit":{"parents":[{"commit":"9521696091e66cbc9190ed7edc4a898ec1c34ff1","subject":"Join do_setup and check_for_setup_error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9521696091e66cbc9190ed7edc4a898ec1c34ff1"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-10-27 13:09:00.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-04-06 10:11:02.000000000","tz":120},"subject":"Docs: Document clone_image driver method","message":"Docs: Document clone_image driver method\n\nThis patch improves and fixes the documentation of the clone_image\ndriver method and updates the parameters in various places where the\nmethod is declared where they presented the wrong parameters.\n\nSince there is a valid default implementation for the method, a small\ntext has been added to indicate that some of the core methods, like this\none, are optional.\n\nTo avoid getting docstrings from the driver and the interface (which is\nthe one used for the documentation) out of sync we replace the\ndescription on cinder/volume/driver.py with a text referencing the\ninterface docstring.\n\nChange-Id: I09a8357e15b9041790285ab1f9a3cdd7a18243b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08b75f26e6484c4461ea1e1be0cf081687665e25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08b75f26e6484c4461ea1e1be0cf081687665e25"}]},"branch":"refs/heads/master"},"16a5fce1dbc98ef36dab635f347f93cfe5d605bb":{"kind":"REWORK","_number":7,"created":"2022-04-06 15:28:00.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/60/815660/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/815660/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/815660/7"}}},"commit":{"parents":[{"commit":"9ca667ecd14dbb9b5ad6bbbb9c8e28b9e0b26683","subject":"Join do_setup and check_for_setup_error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ca667ecd14dbb9b5ad6bbbb9c8e28b9e0b26683"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-10-27 13:09:00.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-04-06 15:04:46.000000000","tz":120},"subject":"Docs: Document clone_image driver method","message":"Docs: Document clone_image driver method\n\nThis patch improves and fixes the documentation of the clone_image\ndriver method and updates the parameters in various places where the\nmethod is declared where they presented the wrong parameters.\n\nSince there is a valid default implementation for the method, a small\ntext has been added to indicate that some of the core methods, like this\none, are optional.\n\nTo avoid getting docstrings from the driver and the interface (which is\nthe one used for the documentation) out of sync we replace the\ndescription on cinder/volume/driver.py with a text referencing the\ninterface docstring.\n\nChange-Id: I09a8357e15b9041790285ab1f9a3cdd7a18243b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/16a5fce1dbc98ef36dab635f347f93cfe5d605bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/16a5fce1dbc98ef36dab635f347f93cfe5d605bb"}]},"branch":"refs/heads/master"},"55e1787dcd6f7f2a4b64c993a0c37161fb2c3478":{"kind":"REWORK","_number":8,"created":"2022-04-08 10:13:32.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/60/815660/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/815660/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/815660/8"}}},"commit":{"parents":[{"commit":"9ca667ecd14dbb9b5ad6bbbb9c8e28b9e0b26683","subject":"Join do_setup and check_for_setup_error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ca667ecd14dbb9b5ad6bbbb9c8e28b9e0b26683"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-10-27 13:09:00.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-04-08 10:13:16.000000000","tz":120},"subject":"Docs: Document clone_image driver method","message":"Docs: Document clone_image driver method\n\nThis patch improves and fixes the documentation of the clone_image\ndriver method and updates the parameters in various places where the\nmethod is declared where they presented the wrong parameters.\n\nSince there is a valid default implementation for the method, a small\ntext has been added to indicate that some of the core methods, like this\none, are optional.\n\nTo avoid getting docstrings from the driver and the interface (which is\nthe one used for the documentation) out of sync we replace the\ndescription on cinder/volume/driver.py with a text referencing the\ninterface docstring.\n\nChange-Id: I09a8357e15b9041790285ab1f9a3cdd7a18243b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/55e1787dcd6f7f2a4b64c993a0c37161fb2c3478"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/55e1787dcd6f7f2a4b64c993a0c37161fb2c3478"}]},"branch":"refs/heads/master"},"e071ee263201d5ed0689e0755cf06f0a8568d16c":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-05-12 18:03:06.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/60/815660/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/815660/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/815660/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/815660/9"}}},"commit":{"parents":[{"commit":"eb6ba624733655b267e1f8d17f0a479007727ff7","subject":"Merge \"mypy: annotate remotefs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb6ba624733655b267e1f8d17f0a479007727ff7"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-10-27 13:09:00.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-05-12 18:02:56.000000000","tz":120},"subject":"Docs: Document clone_image driver method","message":"Docs: Document clone_image driver method\n\nThis patch improves and fixes the documentation of the clone_image\ndriver method and updates the parameters in various places where the\nmethod is declared where they presented the wrong parameters.\n\nSince there is a valid default implementation for the method, a small\ntext has been added to indicate that some of the core methods, like this\none, are optional.\n\nTo avoid getting docstrings from the driver and the interface (which is\nthe one used for the documentation) out of sync we replace the\ndescription on cinder/volume/driver.py with a text referencing the\ninterface docstring.\n\nChange-Id: I09a8357e15b9041790285ab1f9a3cdd7a18243b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e071ee263201d5ed0689e0755cf06f0a8568d16c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e071ee263201d5ed0689e0755cf06f0a8568d16c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
