)]}'
{"id":"openstack%2Fcinder~816324","triplet_id":"openstack%2Fcinder~master~I12018c3a1f6e4e59a4923ef81da85c2d7c74e96d","project":"openstack/cinder","branch":"master","topic":"bug/1949061","attention_set":{},"removed_from_attention_set":{"32036":{"account":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"last_update":"2021-11-12 05:00:10.000000000","reason":"katari manoj kumar replied on the change"},"34201":{"account":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"last_update":"2021-12-13 16:18:47.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2021-12-13 14:49:41.000000000","reason":"Eric Harney replied on the change"}},"hashtags":[],"change_id":"I12018c3a1f6e4e59a4923ef81da85c2d7c74e96d","subject":"[SVF] Fix Retype issue of mirror volume","status":"MERGED","created":"2021-11-02 11:54:08.000000000","updated":"2021-12-13 16:21:35.000000000","submitted":"2021-12-13 16:18:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"816324-bug/1949061","meta_rev_id":"f42c698c508be98ebbf43c558610586c1ab9644b","_number":816324,"virtual_id_number":816324,"owner":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-13 16:18:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"value":2,"date":"2021-12-13 14:49:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-13 14:56:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-13 14:56:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-02 12:23:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-02 12:43:43.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-02 13:00:18.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-02 13:01:30.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-02 13:06:05.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-11-02 13:06:30.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-11-02 13:15:56.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-02 14:15:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-02 14:18:40.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-11-02 15:27:13.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-02 16:03:47.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-02 18:04:48.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-03 04:41:02.000000000","updated_by":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"reviewer":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"state":"REVIEWER"},{"updated":"2021-11-03 11:46:14.000000000","updated_by":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"reviewer":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"state":"CC"},{"updated":"2021-11-05 05:45:14.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2021-11-05 07:01:04.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-11-05 08:33:06.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-11-05 08:40:11.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-05 09:07:02.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-11-08 07:47:22.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-11-08 10:11:02.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-11-08 13:01:18.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-11-19 20:22:19.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-20 00:01:45.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-20 00:48:26.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-21 21:36:47.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-12-13 14:25:47.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-12-13 14:56:49.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"caf3186edc7b60c7f65e7312ba0754ea2d362441","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2021-11-02 11:54:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"534cb4c30c8bf142d2f860bd3dd0e555277f493b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 12:23:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/001417ee55c148959482761f2fdabb79 : SUCCESS in 22m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00bcead6101247bdb2f6295d0f0341ea : SUCCESS in 27m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9764f177ac0bc97332370d55576e18696a0a87b0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-02 12:43:43.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"82d91d1456d04fdb68aa95344befcbaf3ed87c6f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-02 13:00:18.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/24/816324/1/check/DellEMC_XtremIO_FC/56d30c1/DellEMC_XtremIO_FC/66 : SUCCESS in 1h 05m 46s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/24/816324/1/check/DellEMC_XtremIO_iSCSI/1489e96/DellEMC_XtremIO_iSCSI/92 : FAILURE in 29m 04s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"c3622edba6190ac53be069840583dc91e166553a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-02 13:01:30.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/24/816324/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"91eaf8ea88b64e542f0620c832aaf357bbaa82ff","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-02 13:06:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/24/816324/1/check/dsvm-tempest-cisco-zonemanager-job/904d0ce/ : SUCCESS in 1h 06m 42s","accounts_in_message":[],"_revision_number":1},{"id":"4696d6daf3487914ddac077925d28d8e895e5736","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-02 13:06:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/816324/1/iscsi/ : SUCCESS in 1h 11m 35s\n- cinder-smb http://cloudbase-ci.com/cinder/816324/1/smb/ : SUCCESS in 51m 21s","accounts_in_message":[],"_revision_number":1},{"id":"b0188b48140735e9159aa5f1e418ed236ee4439c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-02 13:15:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/24/816324/1/check/iscsi-driver-primera/2c1e1e7/2c1e1e7.html : SUCCESS in 1h 19m 21s","accounts_in_message":[],"_revision_number":1},{"id":"7265da0c100e6a52cc993f6aed9d748f32ed31b6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-02 13:38:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/24/816324/1/check/iscsi-driver-alletra-6k/4fbeb8d/4fbeb8d.html : SUCCESS in 1h 42m 02s","accounts_in_message":[],"_revision_number":1},{"id":"34384466dce0758def38814b709834795ec771a4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-02 13:47:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/24/816324/1/check/iscsi-driver-nimble/dc4a291/dc4a291.html : SUCCESS in 1h 51m 10s","accounts_in_message":[],"_revision_number":1},{"id":"8048a587eb3829af881fe2032adec6c5f31e1306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 14:15:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adc2dc573185417fb88f751598189a89 : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6eb6332177854a9dae779f83ad9180d7 : SUCCESS in 11m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/937b79ae275c4dbc969dccd1ea2bd6c0 : SUCCESS in 9m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/397b00548adf427f9fde5ae204c11a16 : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9671b40a01dd40dcba7cbe4bbb60d375 : SUCCESS in 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/9b4ee9408766490b9785914f333cb587 : SUCCESS in 56m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1ed7ae0f1bd74dddbd80331e037a6815 : SUCCESS in 1h 22m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fa25f4f260cc421db873caf44dc97013 : SUCCESS in 13m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/089872157dc04468a152ef5fe56562f1 : SUCCESS in 6m 33s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/41205029de174b8a93ecfeeff9c30cea : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e6fa709e66e747f987ac55ce33a98ce2 : SUCCESS in 7m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ea0a3965e5e54ea9b6717b8b75a2b178 : SUCCESS in 6m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f67431da02b84296afe869752668895a : SUCCESS in 35m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/593bf0a5de7641fa937be94e5aad0687 : FAILURE in 5m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/341c42a60fc84561a9d9990036b97947 : SUCCESS in 1h 42m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c054c62934244109b1f0cc1da260b72a : SUCCESS in 1h 46m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e72ea7ba33624dbaad4552169db7ff52 : FAILURE in 1h 50m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6ad8a53e122b484c932eb0f40af004e8 : SUCCESS in 1h 05m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b84a276d8bed48418be35c10036712bb : SUCCESS in 1h 06m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/04aa9222aff14f8fa5a40810c9227c17 : SUCCESS in 31m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/477624efad2e46b2855b45ad1aaec32c : SUCCESS in 1h 20m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c783668760ad4599989423cf90a2c40b : FAILURE in 2h 18m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0881540ed71d46e8b2a70a63001c2435 : SUCCESS in 48m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ceab48be70e34e85b53999264c6cba3a : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ae3a9f8955ff8aef2f45a5396a4b3fe0428e403a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-02 14:18:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/24/816324/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/19e4824/ : SUCCESS in 2h 23m 29s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/24/816324/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/4e37356/ : SUCCESS in 1h 36m 06s","accounts_in_message":[],"_revision_number":1},{"id":"363f1345572fd4b217d26c0ad9660bb82895b7c2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-02 15:11:03.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/24/816324/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"9b15f95f3d038774cfb0bfb938a9082a83f3dc6f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-02 15:27:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/816324/1/20211102205411/NEC-ISCSI : SUCCESS in 3h 21m 11s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/816324/1/20211102205411/NEC-FC : SUCCESS in 3h 32m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e51ac8365e603123cf11e65bc1d0ba1daef3bc91","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-02 16:03:47.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-816324-1/ns5_nfs : SUCCESS in 4h 5m 28s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-816324-1/ns5_iscsi : SUCCESS in 4h 8m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"55a4fa07f613630b7549fd4e1761a6c01523d912","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-02 17:20:42.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/24/816324/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"6f6cb7dd73fe28c612ab176803497bd50dcc0ecb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-11-02 18:04:48.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/816324/816324/1/check-not-vote/ext-cinder-zuul/abff641 : SUCCESS in 36m 38s","accounts_in_message":[],"_revision_number":1},{"id":"afec5a1a571ccf4f0d4723456357800d46633461","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2021-11-03 11:40:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"af67f0491d1a190d656f9d801c52c0fdfeac3ed1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2021-11-03 11:46:14.000000000","message":"Patch Set 2:\n\nBuild Failed. ibm-gpfs-dsvm-tempest-full: s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/90/index.html","accounts_in_message":[],"_revision_number":2},{"id":"44d03403366ab28aefe84ba3ebf4e28de6f8d263","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 12:07:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/930aada31357441e9b2b59293f9695a4 : SUCCESS in 18m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/61b21d4015ac49459f8be679da3910f0 : SUCCESS in 24m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0eb06d19e5a9394fac6e75654f5ed647c727ad86","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-03 12:12:46.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/24/816324/2/check/iscsi-driver-alletra-6k/e909284/e909284.html : FAILURE in 30m 13s","accounts_in_message":[],"_revision_number":2},{"id":"a878da52cd2004ab861b6a90401449018550a299","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-03 12:44:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/816324/2/iscsi/ : SUCCESS in 1h 03m 28s\n- cinder-smb http://cloudbase-ci.com/cinder/816324/2/smb/ : SUCCESS in 51m 40s","accounts_in_message":[],"_revision_number":2},{"id":"9e18233dff7e3c7ccf9f1ff524cc1c3df0c6427e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-03 12:51:02.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/24/816324/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ceed3ed0138b7966b8d43c5447087532ec34f9e0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-03 12:55:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/24/816324/2/check/iscsi-driver-nimble/01ce1fc/01ce1fc.html : FAILURE in 1h 13m 29s","accounts_in_message":[],"_revision_number":2},{"id":"a9ebf409f39ad407323450902e07ca81c3bde701","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-03 12:58:55.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/24/816324/2/check/dsvm-tempest-cisco-zonemanager-job/547f212/ : SUCCESS in 1h 07m 59s","accounts_in_message":[],"_revision_number":2},{"id":"4ffe4b64c88472fff1461bdb7c4f7623b5a48a23","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-03 13:03:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/24/816324/2/check/iscsi-driver-primera/7831b2a/7831b2a.html : SUCCESS in 1h 20m 42s","accounts_in_message":[],"_revision_number":2},{"id":"41619a2687cde0c3e3ad5e3b67c75b5c12d15d1d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-03 13:53:33.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/24/816324/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4f3cf35/ : SUCCESS in 2h 11m 46s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/24/816324/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f686eff/ : SUCCESS in 1h 58m 24s","accounts_in_message":[],"_revision_number":2},{"id":"eacedcc3437d991a5b6a78fa716d6b16fe650e4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 14:03:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9997aebc86d9461f930c0006762315aa : SUCCESS in 9m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41863d0145fa4a18b98e8808b634a450 : SUCCESS in 13m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee1520e554b048c2a19c387a4ae44a80 : TIMED_OUT in 41m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d157296c4d5645c790ca18f2d37358cd : SUCCESS in 13m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7778d12fc7fb4919a7b7c798557718f6 : SUCCESS in 15m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/7bf6232d227e4d96b542bc180f0505ff : SUCCESS in 57m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5c614e7289184b778b116a2757b1c94d : SUCCESS in 1h 13m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad20b03100a14bb6ba89aeb4c0d12be9 : SUCCESS in 8m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/28bf6e4b98644ebdb15ffd894705dabd : SUCCESS in 11m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6b5658c178c046e181b9eb703d067cd3 : SUCCESS in 8m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d31f1e4a8306468383ea5fe3245e21e9 : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/36d7bfe00a7941988d59e61be9ea3988 : SUCCESS in 8m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3ffd34bd531e40d3971d63266b1dde6b : SUCCESS in 6m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2a0a010e8c764b788ac76ed02b68e265 : SUCCESS in 54m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e677369f214344a6b57b60dcdd18fd7b : FAILURE in 7m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0071f3dfa6ac4332ab67240a17f85358 : SUCCESS in 1h 12m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bf6f64734fed423993876c6135b1e989 : SUCCESS in 1h 45m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/db31916ea40a47da84fc6aa55a8e0318 : SUCCESS in 1h 38m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7d8e6d6f1a6b44998350d652d7ba12fa : SUCCESS in 1h 08m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c2eadb9cdf5d4b86bbe60ca1b6acf01e : SUCCESS in 1h 11m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5849168e552e48218e47fa36fcb947ab : SUCCESS in 39m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4af3916217a742d2904a96890b974955 : SUCCESS in 1h 28m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d1232ce04643407285d1b894cfece2a0 : SUCCESS in 2h 16m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/becca96921664cf098515ed641c85226 : SUCCESS in 50m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11c7f5bb202a4094b82081f61a9fe1de : SUCCESS in 54m 01s","accounts_in_message":[],"_revision_number":2},{"id":"b3341b70a601860e8c21aa156a5f1a3ccb59e379","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-03 14:15:25.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/24/816324/2/check/tempest-dsvm-full-focal-py3/ed46ed4/ : FAILURE in 2h 29m 57s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/24/816324/2/check/devstack-plugin-ceph-dsvm-focal-py3/2683b8a/ : FAILURE in 35m 38s","accounts_in_message":[],"_revision_number":2},{"id":"8d61715a5421e44d28f362356684eb85706357cd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-03 15:34:02.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/24/816324/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"eb330beb1eec4fcf23a70f344775a4221904a95c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-03 15:45:54.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-816324-2/ns5_nfs : SUCCESS in 4h 1m 34s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-816324-2/ns5_iscsi : SUCCESS in 4h 4m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"cdeb3e2b8eb0006fefe4376383cdf1a22fcb9c99","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-11-03 15:48:24.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/816324/816324/2/check-not-vote/ext-cinder-zuul/016e64f : SUCCESS in 32m 18s","accounts_in_message":[],"_revision_number":2},{"id":"65a31fff7b123e3f859e34ae985ec1eb905e4f0d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-03 16:58:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/816324/2/20211103220917/NEC-FC : SUCCESS in 3h 31m 18s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/816324/2/20211103220917/NEC-ISCSI : SUCCESS in 3h 49m 28s","accounts_in_message":[],"_revision_number":2},{"id":"9a47f692ab995e2ab82b1886c738ef0faadaa66e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-03 17:01:49.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/24/816324/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"0a64acb9f60a4a0fec6f459c820003fa0a7e6bcf","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2021-11-05 04:42:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b2200c8396184fae16a8de9c922b5951fc8daa68","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-05 04:57:07.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/816324/2/iscsi/ : FAILURE in 13m 54s\n- cinder-smb http://cloudbase-ci.com/cinder/816324/2/smb/ : FAILURE in 13m 43s","accounts_in_message":[],"_revision_number":2},{"id":"5434005c53eeaa6c93a1e51b112553e7d3ffe08c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 05:10:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b41d49913db4cd0835998f5d0afb75f : SUCCESS in 18m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bbbdfe328a5f4f7cbd7d89b9daf87d1b : SUCCESS in 21m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f4619302f8be82c56e09f143e47e4b3f0875667d","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-11-05 05:45:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/cbe146c/index.html : SUCCESS in 1h 01m 38s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1d330da/index.html : SUCCESS in 1h 01m 54s","accounts_in_message":[],"_revision_number":2},{"id":"fda26f52b62a70945babdaac18f95933d28b3bdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 06:37:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee2f55bc42c947c39c02edcd1ad87ebe : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ca8f964e7d54d688abd4971fb8001c7 : SUCCESS in 10m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea00ccc808aa457bb67a7f234622d856 : SUCCESS in 9m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcca8df7ebaf43f9af9cbee65cb3dc1f : SUCCESS in 11m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ded22c467498476a9bd2b18dd8482525 : SUCCESS in 10m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/d44a58f6906d4df58ce5b9b945659487 : SUCCESS in 52m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fefe02d1bc40453cb5db140408927602 : SUCCESS in 1h 23m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44c92cee7dab4c5281f4db52d298256c : SUCCESS in 7m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8a07cdc3cf7a4e16895673d46d29c106 : SUCCESS in 13m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/352572cbc3174a3392f17dbf11f9c856 : SUCCESS in 6m 05s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/af930f914a3746579e3a9ef0ed690b58 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/abb0bac2174d4f7c9e7f7b3dd51b3089 : SUCCESS in 7m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/af2ceeefeebd43f2a5d375181fd4a146 : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fd431e544839460ea4407af329d89ca4 : FAILURE in 26m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/91ccbe3e435e4fe199c7c49770053a4b : FAILURE in 4m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ddbb9f5bc5fc4bc2b84e7773f1e7a062 : SUCCESS in 1h 19m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1484e9f354f54d00813bb2ba44c5defa : SUCCESS in 1h 25m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/80e6fefc9a494093bf9495c2f2206a31 : SUCCESS in 1h 22m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/38e617d3210b4da5b2e1f446531f2d76 : SUCCESS in 1h 19m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/26a126b4a1854bad97fb2f1a21f12a4a : SUCCESS in 1h 01m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/61446858d6044ae8817242f56ff5bca3 : SUCCESS in 44m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/723b0daf793443269fb5f9428de5c801 : SUCCESS in 1h 14m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c019c973427b47ce8696b2c90cfddae2 : SUCCESS in 1h 47m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/064e36e9444449a792e76b947e6c8c45 : SUCCESS in 58m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5577f270fb45455c990e66e4680dc150 : SUCCESS in 43m 15s","accounts_in_message":[],"_revision_number":2},{"id":"26c869187e85bb8c3825b73c3a4dc3eb806311fa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-05 06:43:23.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/24/816324/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"0e880b7f378aaeeda8dab9767091e97e1e5bc02e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-11-05 06:45:39.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/816324/816324/2/check-not-vote/ext-cinder-zuul/2c6acb3 : FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"8f1182e5d967f3b512aca9bc2831ddbce16d7bca","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-05 06:54:44.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/24/816324/2/check/devstack-plugin-ceph-dsvm-focal-py3/fa0c0cc/ : FAILURE in 1h 12m 40s","accounts_in_message":[],"_revision_number":2},{"id":"cf795c5544f4db313582cce2d2189826c189d6ac","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-11-05 07:01:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/24/816324/2/upstream-solidfire-check/cinder-solidfire-iSCSI/f248588/ : FAILURE in 22m 17s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"6c6301a74b52dd9ff85e095299bfeff651444ca6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-11-05 08:33:06.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/24/816324/2/check/DellEMC_PowerStore_iSCSI/f74f13b/DellEMC_PowerStore_iSCSI/38 : SUCCESS in 1h 55m 06s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/24/816324/2/check/DellEMC_PowerStore_FC/eaf8af1/DellEMC_PowerStore_FC/45 : FAILURE in 1h 25m 38s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"7f11d6a085825b6df2fca2d7c1b3ad68109f1bc4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-05 08:40:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/24/816324/2/upstream-check/cinder-cDOT-iSCSI/40d76b5/ : SUCCESS in 2h 01m 38s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/24/816324/2/upstream-check/cinder-cDOT-NFS/c87a461/ : SUCCESS in 1h 56m 13s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/24/816324/2/upstream-check/cinder-cDOT-FCP/c7976bc/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"1c24ed0195da33aac64c5ee33054f32d435ddf8d","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-11-05 09:07:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI http://os-logs.tristero.net/dell-ci/816324-2/PVME-iSCSI : SUCCESS in 58:18.49\n- PVME-FC http://os-logs.tristero.net/dell-ci/816324-2/PVME-FC : SUCCESS in 1:00:06","accounts_in_message":[],"_revision_number":2},{"id":"9880e939ebf5b7d1a2d1563b401b95ab4986a93f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-05 10:20:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/24/816324/2/check/dsvm-tempest-cisco-zonemanager-job/93edbbd/ : SUCCESS in 1h 07m 57s","accounts_in_message":[],"_revision_number":2},{"id":"32e4ca6f15577763616600fe616a829d915af5bf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-05 10:44:08.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/24/816324/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"61a4de5048b8c927803e550f49fdef871ff56b71","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-11-08 07:47:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/816324-2/Seagate-iSCSI : SUCCESS in 1:20:34\n- Seagate-FC http://os-logs.tristero.net/seagate-ci/816324-2/Seagate-FC : SUCCESS in 47:53.96","accounts_in_message":[],"_revision_number":2},{"id":"03ec57add0d0f408dd76577599f8da93ee645d4c","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-11-08 10:11:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/816324-2/HPMSA-iSCSI : SUCCESS in 1:11:25\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/816324-2/HPMSA-FC : SUCCESS in 44:04.99","accounts_in_message":[],"_revision_number":2},{"id":"c51d7bdffb0d9902f677b2509f4aa2ffc332a7a0","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-11-08 13:01:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/816324-2/Lenovo-iSCSI : SUCCESS in 1:24:46\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/816324-2/Lenovo-FC : SUCCESS in 57:10.30","accounts_in_message":[],"_revision_number":2},{"id":"b77501cb381163a176f36f870c63358101608571","author":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"date":"2021-11-12 05:00:10.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c69cd7cdf4c06a35e61d9d6b4bf4a3833c6a07a7","author":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"date":"2021-11-12 05:00:17.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"69164bfbfde03808e2a13c739e940266f1faab9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2021-11-19 12:48:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5789b476068859b197b152dd473491e92880aad3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2021-11-19 12:52:09.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"da720457497387fe6bc96584ed65cc3ce8d238d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 13:22:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46b14129cc3c4b159c3a7e0327676e4b : SUCCESS in 17m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/814af99423ad4c6d8834d9a6777ae570 : FAILURE in 22m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dd5ce843242fd01c0aaf654a431e55a1f75bd14c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-19 13:51:19.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/816324/4/iscsi/ : SUCCESS in 58m 39s\n- cinder-smb http://cloudbase-ci.com/cinder/816324/4/smb/ : SUCCESS in 48m 48s","accounts_in_message":[],"_revision_number":4},{"id":"8135c4f0f45c564fe68ee03e0c963f1b671f2ab1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-19 13:58:37.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/24/816324/4/check/iscsi-driver-alletra-6k/b696f07/b696f07.html : FAILURE in 1h 04m 35s","accounts_in_message":[],"_revision_number":4},{"id":"f3831e21f559835becdcc5e614c541814a4a7f17","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-19 14:08:09.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/24/816324/4/check/iscsi-driver-nimble/9781630/9781630.html : FAILURE in 1h 11m 05s","accounts_in_message":[],"_revision_number":4},{"id":"e0714b6f60bf72b6dcfc34f875c024b6d6af3d55","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-19 14:08:59.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/24/816324/4/check/dsvm-tempest-cisco-zonemanager-job/659d0f6/ : SUCCESS in 1h 10m 01s","accounts_in_message":[],"_revision_number":4},{"id":"3193cb5282a2d9f162730d64f75246965e51ef72","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-19 14:12:46.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/24/816324/4/check/iscsi-driver-primera/136b67e/136b67e.html : SUCCESS in 1h 18m 17s","accounts_in_message":[],"_revision_number":4},{"id":"e112438919445bc0f3ba529f55291dee155d6217","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-19 14:15:45.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/24/816324/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/675e8f9/675e8f9.html : SUCCESS in 1h 20m 59s","accounts_in_message":[],"_revision_number":4},{"id":"f3851038d80cc52b812c59f1fe8f435b9198f932","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-19 14:30:52.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/24/816324/4/check/DellEMC_XtremIO_FC/4f022fa/DellEMC_XtremIO_FC/172 : SUCCESS in 54m 21s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/24/816324/4/check/DellEMC_XtremIO_iSCSI/b4382d4/DellEMC_XtremIO_iSCSI/196 : SUCCESS in 1h 35m 39s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"df680ff365ddfdfc0ed48189ac46ef426ee5d8fe","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-11-19 14:35:20.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/816324/816324/4/check-not-vote/ext-cinder-zuul/05a5157 : SUCCESS in 35m 01s","accounts_in_message":[],"_revision_number":4},{"id":"d37018c527b6db451b33236ba8315f7f0ece2299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 15:17:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f386ad82360346aaab280c09588405d8 : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d35aa3247ae14761928649253c1abea7 : SUCCESS in 16m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34ba109c157e433e878215d1ba05b5bc : SUCCESS in 10m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d37a7e24a1834aa58998317ecde1916b : SUCCESS in 12m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0c76354d9734be0815fcb76301c238e : SUCCESS in 12m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/5c87bf296acb434ebb21960384b39f1b : SUCCESS in 1h 03m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/33322b3605a44357b697ff3154bcba19 : SUCCESS in 1h 03m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b09458c3867c4d1fb1020ddd326c59be : SUCCESS in 7m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1bfbbe9b05684ad58ecb665f2aba9a80 : SUCCESS in 12m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/da21d6825bc24155a3ec0040976b0930 : SUCCESS in 7m 26s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/22e5429be4c74969a788c27520447bc2 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4908fc4c29074dc7b81d268dfa9240d0 : SUCCESS in 7m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c8880fe6a3c04b5eb420bfdcc80c243b : SUCCESS in 6m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e68125eb727c402987a481fd4b4dab84 : SUCCESS in 38m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e16eb551bb0a4c7a8093f6a78ff7d865 : FAILURE in 5m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5d611abf9f8143c49caacd0755cfe4d9 : SUCCESS in 1h 34m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9230dae9d5c0444697188634a6b2dc00 : SUCCESS in 1h 16m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/87575e055a92476bbb8db2a261fcee12 : SUCCESS in 1h 34m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f17f501429cc42dfb3c4d294898dec48 : SUCCESS in 57m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/48ff01ddfa264797bca6d1031c76f236 : SUCCESS in 1h 21m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/73de58326b8d40898d9baaed9d091007 : SUCCESS in 40m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/11ee9fc6844543c7bbcf92fbdbed53e6 : SUCCESS in 1h 10m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/82b5f07df86c46d8896dddbbbc0899af : SUCCESS in 2h 20m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/778ad63e0f4c477f957b606cc4994a8d : SUCCESS in 49m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5055e0145749462ca2b4b1e604e566ed : SUCCESS in 41m 36s","accounts_in_message":[],"_revision_number":4},{"id":"846297184a4d714abce69eda570c83630f2185c7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-19 15:36:09.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/24/816324/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5685817/ : SUCCESS in 2h 16m 13s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/24/816324/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/43081fa/ : FAILURE in 2h 42m 58s","accounts_in_message":[],"_revision_number":4},{"id":"26a6d0cb2d33dbb3701113e21a7dc31fda667f5e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-19 17:14:57.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-816324-3/ns5_nfs : SUCCESS in 4h 25m 11s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-816324-3/ns5_iscsi : SUCCESS in 4h 26m 7s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"0b75caf8b34ea3a907fe5068909ba996aa492b28","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-19 17:40:01.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/24/816324/4/check/tempest-dsvm-full-focal-py3/4e8fd2a/ : SUCCESS in 2h 19m 29s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/24/816324/4/check/devstack-plugin-ceph-dsvm-focal-py3/5c18e27/ : SUCCESS in 2h 52m 54s","accounts_in_message":[],"_revision_number":4},{"id":"70b8cfc0ab9960f08831efeb030f33e6ef915fce","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-19 17:44:38.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/24/816324/4/upstream-check/cinder-cDOT-iSCSI/6b8576b/ : SUCCESS in 32m 47s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/24/816324/4/upstream-check/cinder-cDOT-NFS/2d7cc90/ : SUCCESS in 2h 25m 45s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/24/816324/4/upstream-check/cinder-cDOT-FCP/66e1e9a/ : FAILURE in 2h 09m 41s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"8b14f93fd1b9990c70fceb9ce29eb03c3101c98e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-19 18:00:02.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-816324-4/ns5_nfs : SUCCESS in 4h 26m 2s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-24-816324-4/ns5_iscsi : SUCCESS in 4h 22m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"ff62e9b455d98001aadf6d300754f679fd0f2788","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-11-19 18:15:57.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/24/816324/4/upstream-solidfire-check/cinder-solidfire-iSCSI/2406a75/ : SUCCESS in 2h 56m 45s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"70ef8eedfe3cc95ff4aef4cbef26dc261b6d3a15","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-19 19:51:27.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/24/816324/4/check/DellEMC_XtremIO_FC/3a597d2/DellEMC_XtremIO_FC/175 : SUCCESS in 1h 02m 03s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/24/816324/4/check/DellEMC_XtremIO_iSCSI/203fabf/DellEMC_XtremIO_iSCSI/199 : SUCCESS in 1h 42m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"5f9fb350d79f149531b8240ab41fb9612d3ac384","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-11-19 20:22:19.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/24/816324/4/check/DellEMC_PowerFlex/ff45153/DellEMC_PowerFlex/114 : SUCCESS in 1h 42m 38s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"ffd570b4a0cb7877cb4d4667ab8206e6b67b2d6d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-11-19 21:10:11.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/24/816324/4/check/DellEMC_PowerStore_iSCSI/db0855a/DellEMC_PowerStore_iSCSI/92 : SUCCESS in 1h 58m 32s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/24/816324/4/check/DellEMC_PowerStore_FC/e487956/DellEMC_PowerStore_FC/102 : SUCCESS in 1h 24m 33s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"4cc5138f9848a70bc1472547c2449d024cb3f9d3","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-11-19 23:04:03.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/816324-4/HPMSA-iSCSI : SUCCESS in 1:11:12\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/816324-4/HPMSA-FC : SUCCESS in 44:12.07","accounts_in_message":[],"_revision_number":4},{"id":"846d307fc7c469afe4b3b2b1e0b97a34fd8540cc","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-11-19 23:52:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/816324-4/Seagate-iSCSI : SUCCESS in 1:17:28\n- Seagate-FC http://os-logs.tristero.net/seagate-ci/816324-4/Seagate-FC : SUCCESS in 49:22.55","accounts_in_message":[],"_revision_number":4},{"id":"7f3f0e64585e61c7904af659107d85877dd3b48a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-11-20 00:01:45.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/24/816324/4/check/DellEMC_VNX_iSCSI/3bd752c/DellEMC_VNX_iSCSI/102 : FAILURE in 2h 58m 15s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/24/816324/4/check/DellEMC_VNX_FC/eb32d69/DellEMC_VNX_FC/103 : SUCCESS in 2h 19m 57s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"6a55dda9f515031a88eddae99d7aec364f535729","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-20 00:33:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/816324/4/20211120064509/NEC-ISCSI : SUCCESS in 2h 42m 21s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/816324/4/20211120064509/NEC-FC : SUCCESS in 2h 47m 52s","accounts_in_message":[],"_revision_number":4},{"id":"4705d237929f4d6ad23187693f4ed33aa6c5271d","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-11-20 00:48:26.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/24/816324/4/check/DellEMC_PowerMAX_FC/cfb5703/DellEMC_PowerMAX_FC/47 : FAILURE in 3h 01m 37s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/24/816324/4/check/DellEMC_PowerMAX_iSCSI/4c58ef2/DellEMC_PowerMAX_iSCSI/46 : FAILURE in 3h 10m 36s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"838ac7e147f743c30d13bb9f88324e1af6a453b1","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-11-20 01:40:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/816324-4/Lenovo-iSCSI : SUCCESS in 1:25:50\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/816324-4/Lenovo-FC : SUCCESS in 1:00:33","accounts_in_message":[],"_revision_number":4},{"id":"0cc4f4da79ca233c0a4f514aef5da99d29581dcc","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-11-21 21:36:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/24/816324/4/check/dsvm-tempest-TOYOU-ACS5000-driver/06aa924 : SUCCESS in 2h 02m 43s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/24/816324/4/check/fc-test/e973fb4 : SUCCESS in 2h 04m 06s","accounts_in_message":[],"_revision_number":4},{"id":"3642d05423d16689b1229e0c07ecf0a6a14d38ba","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2021-11-22 05:18:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f818e7e7bbc10f438c527d8057068b8a03b19cee","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-11-22 06:24:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1813e76/index.html : SUCCESS in 1h 04m 16s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/54affaf/index.html : SUCCESS in 1h 05m 55s","accounts_in_message":[],"_revision_number":4},{"id":"289319c2497a0edbae9afefa81adcefbc34c23e0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2021-11-30 17:17:48.000000000","message":"Patch Set 3:\n\nBuild Failed. ibm-gpfs-dsvm-tempest-full: s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/151/index.html","accounts_in_message":[],"_revision_number":3},{"id":"aab60ae0a06d4d7cf4d59c5f3b727ee2b7899aca","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-12-13 14:49:41.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e146d81468ebc3af1a1c2460269f63439bd6ccf7","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-12-13 14:56:49.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8037303c51070fdf744d5564184dfd3212e1c0e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 14:57:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c0e3364ee6fc4df3ac31bea50bd6697f5721d0bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 16:18:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24c0ca2281514d35b1761c2b1c0b466a : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1be765b2d1d743b6955f08997944eeb3 : SUCCESS in 12m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/219c20140491450084a187eb4112998d : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b8284f98a1d4518abe974e7f67dac5b : SUCCESS in 13m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/190254c10dc0417cad8810fe02207afd : SUCCESS in 58m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a99d5819746e454c8d606589b0e77619 : SUCCESS in 1h 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3770f3f4f9a4d3aa17305cbfc7cc494 : SUCCESS in 46m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b0c90ef9f804c4a865cf85024ff046d : SUCCESS in 7m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c79fb51ac7fd474994972ec684f44536 : SUCCESS in 1h 06m 17s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a30ae731e8b94315aeb40fc19c181390 : SUCCESS in 1h 20m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23da1090160d4eea8e25ee769247c859 : SUCCESS in 1h 04m 05s","accounts_in_message":[],"_revision_number":4},{"id":"506790a8cc6135fa8afe48bb2cd602c2a89797b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 16:18:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f42c698c508be98ebbf43c558610586c1ab9644b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 16:21:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e63b1c8030804229aaa83c7e0c56e0da : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41e4054f932b4a9c938e556f9546bbb1 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"50e173101162b32949bc51eba288c5b773cb9cb1","revisions":{"e1354c3795e48ea71936711a2fe6193b0b12d1be":{"kind":"REWORK","_number":1,"created":"2021-11-02 11:54:08.000000000","uploader":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"ref":"refs/changes/24/816324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/816324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/816324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/816324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/816324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/816324/1"}}},"commit":{"parents":[{"commit":"aafaba0e95b02cdb7a0f00401385bfc8f723449d","subject":"Merge \"Update rbd warning message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aafaba0e95b02cdb7a0f00401385bfc8f723449d"}]}],"author":{"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","date":"2021-11-02 10:23:02.000000000","tz":0},"committer":{"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","date":"2021-11-02 10:23:02.000000000","tz":0},"subject":"[SVF] Fix Retype issue of mirror volume","message":"[SVF] Fix Retype issue of mirror volume\n\n[Spectrum Virtualize Family] Fixed retype issue of\nnon-rep mirror volume to mirror-volume-type with different\nmirror pool.\n\nCloses-Bug: #1949061\nChange-Id: I12018c3a1f6e4e59a4923ef81da85c2d7c74e96d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e1354c3795e48ea71936711a2fe6193b0b12d1be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e1354c3795e48ea71936711a2fe6193b0b12d1be"}]},"branch":"refs/heads/master"},"f751c64f76d33386a1367efedae5b477cda0e7a5":{"kind":"REWORK","_number":2,"created":"2021-11-03 11:40:45.000000000","uploader":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"ref":"refs/changes/24/816324/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/816324/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/816324/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/816324/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/816324/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/816324/2"}}},"commit":{"parents":[{"commit":"aafaba0e95b02cdb7a0f00401385bfc8f723449d","subject":"Merge \"Update rbd warning message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aafaba0e95b02cdb7a0f00401385bfc8f723449d"}]}],"author":{"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","date":"2021-11-02 10:23:02.000000000","tz":0},"committer":{"name":"sreerammounika","email":"sreeram.mounika@ibm.com","date":"2021-11-03 11:39:34.000000000","tz":0},"subject":"[SVF] Fix Retype issue of mirror volume","message":"[SVF] Fix Retype issue of mirror volume\n\n[Spectrum Virtualize Family] Fixed retype issue of\nnon-rep mirror volume to mirror-volume-type with different\nmirror pool.\n\nCloses-Bug: #1949061\nChange-Id: I12018c3a1f6e4e59a4923ef81da85c2d7c74e96d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f751c64f76d33386a1367efedae5b477cda0e7a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f751c64f76d33386a1367efedae5b477cda0e7a5"}]},"branch":"refs/heads/master"},"ddaf0c2782ddc69921b91cf81cedf4506dc3b83a":{"kind":"REWORK","_number":3,"created":"2021-11-19 12:48:07.000000000","uploader":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"ref":"refs/changes/24/816324/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/816324/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/816324/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/816324/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/816324/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/816324/3"}}},"commit":{"parents":[{"commit":"aafaba0e95b02cdb7a0f00401385bfc8f723449d","subject":"Merge \"Update rbd warning message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aafaba0e95b02cdb7a0f00401385bfc8f723449d"}]}],"author":{"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","date":"2021-11-02 10:23:02.000000000","tz":0},"committer":{"name":"sreerammounika","email":"sreeram.mounika@ibm.com","date":"2021-11-19 12:40:07.000000000","tz":0},"subject":"[SVF] Fix Retype issue of mirror volume","message":"[SVF] Fix Retype issue of mirror volume\n\n[Spectrum Virtualize Family] Retype of mirror-volume to volume-type\nwith different mirror-pool scenario is not working.\n\nWhen there is a change of mirror pool in retype operation then it is\nchecking whether source/target type has data reduction pool. If it\ncontains data reduction pool, it will throw an exception. If not,\nit has to continue the retype operation with change of mirror pool.\nDue to the incorrect place of raising an exception for\ndata-reduction pool,it is effecting the retype of change in\nmirror-pool operation.\n\nUpdated the code to raise an exception correctly to fix this issue.\n\nCloses-Bug: #1949061\nChange-Id: I12018c3a1f6e4e59a4923ef81da85c2d7c74e96d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ddaf0c2782ddc69921b91cf81cedf4506dc3b83a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ddaf0c2782ddc69921b91cf81cedf4506dc3b83a"}]},"branch":"refs/heads/master"},"50e173101162b32949bc51eba288c5b773cb9cb1":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-11-19 12:52:09.000000000","uploader":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"ref":"refs/changes/24/816324/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/816324/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/816324/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/816324/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/816324/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/816324/4"}}},"commit":{"parents":[{"commit":"aafaba0e95b02cdb7a0f00401385bfc8f723449d","subject":"Merge \"Update rbd warning message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aafaba0e95b02cdb7a0f00401385bfc8f723449d"}]}],"author":{"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","date":"2021-11-02 10:23:02.000000000","tz":0},"committer":{"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","date":"2021-11-19 12:52:09.000000000","tz":0},"subject":"[SVF] Fix Retype issue of mirror volume","message":"[SVF] Fix Retype issue of mirror volume\n\n[Spectrum Virtualize Family] Retype of mirror-volume to volume-type\nwith different mirror-pool scenario is not working.\n\nWhen there is a change of mirror-pool in retype operation then it is\nchecking whether source/target type has data reduction pool. If it\ncontains data reduction pool, it will throw an exception. If not,\nit has to continue the retype operation with change of mirror pool.\nDue to the incorrect place of raising an exception for\ndata-reduction pool,it is effecting the retype operation of\nvolume-type with different mirror-pool.\n\nUpdated the code to raise an exception correctly to fix this issue.\n\nCloses-Bug: #1949061\nChange-Id: I12018c3a1f6e4e59a4923ef81da85c2d7c74e96d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50e173101162b32949bc51eba288c5b773cb9cb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50e173101162b32949bc51eba288c5b773cb9cb1"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
