)]}'
{"id":"openstack%2Fcinder~816695","triplet_id":"openstack%2Fcinder~stable%2Fxena~I5b660776190f3026296d6d3237bd3b0d609f769f","project":"openstack/cinder","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"12670":{"account":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"last_update":"2021-11-18 21:12:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5b660776190f3026296d6d3237bd3b0d609f769f","subject":"PowerMax Driver - Fix for legacy PowerMax OS around generations","status":"MERGED","created":"2021-11-04 13:30:51.000000000","updated":"2021-11-18 21:15:03.000000000","submitted":"2021-11-18 21:12:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816695","meta_rev_id":"cbd302d7a4cfc6e74620d0e21f96fd485c27d20e","_number":816695,"virtual_id_number":816695,"owner":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-18 21:12:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-12 13:17:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2021-11-18 09:26:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2021-11-18 09:26:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"}],"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-04 14:23:53.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-04 14:24:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-04 14:43:48.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-11-04 16:12:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-04 16:29:16.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-11-04 16:35:23.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-11-04 17:44:17.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2021-11-04 19:08:39.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-04 19:33:18.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-04 20:26:07.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-04 20:30:23.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-11-09 12:59:48.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-11-12 13:17:22.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-11-18 09:26:44.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"fa18c01e80e961519415274b271a6fefdce3a795","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-11-04 13:30:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d932d4792cd15af8f456e4efaa3d119112130ca0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-04 14:23:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/95/816695/1/check/iscsi-driver-primera/96633cd/96633cd.html : FAILURE in 51m 05s","accounts_in_message":[],"_revision_number":1},{"id":"fe63d067d078adad3f7c088bf83d55c66e08c197","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 14:24:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aac212c210d44f059f3ce98a87a76eb7 : SUCCESS in 19m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b8e7ce6c12e4c828f3750446112471a : SUCCESS in 23m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd06ba1991dcfd64dac0a48e888645d7aae775ee","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-04 14:24:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/95/816695/1/check/iscsi-driver-alletra-6k/796ef8d/796ef8d.html : FAILURE in 52m 48s","accounts_in_message":[],"_revision_number":1},{"id":"587f9289c1b856b12cf80dda97b6dee3af4d53a6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-04 14:43:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/816695/1/iscsi/ : SUCCESS in 1h 12m 27s\n- cinder-smb http://cloudbase-ci.com/cinder/816695/1/smb/ : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":1},{"id":"272186e4a3cf574cbd05ca3480cfee3952338d84","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-04 15:41:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/95/816695/1/check/iscsi-driver-nimble/71c52c8/71c52c8.html : FAILURE in 1h 00m 27s","accounts_in_message":[],"_revision_number":1},{"id":"8579fc7b06c5cfad06ab1abe73da3405b60d63b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 16:12:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00453df2dc234514bed1d53f6648712b : SUCCESS in 9m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f04c2917719f44c58344cbb9d4321bc6 : SUCCESS in 11m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d636bd2f5ce04eb8b5b20344e906c6c5 : SUCCESS in 11m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c630f19e8c3c442d93743b86df00416e : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d767886c21004a3ea1a6f5372c6bcbae : SUCCESS in 13m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/a9823ce798a7435d9941df330571c58f : SUCCESS in 1h 05m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/76f4279668e44dc09023f7bedb266d67 : SUCCESS in 1h 11m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96ce10fb77524b05b7ed393f7b6cbf7c : SUCCESS in 9m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5448fef1c0f047f899c7f515541e43d9 : SUCCESS in 15m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7fd12ad328cb4f2dbed7760b45caae4b : SUCCESS in 6m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b5f6ebbe52c74a3aa65bbebf69177f3f : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7f5ce01f8340499fb35e79257cf45d95 : SUCCESS in 9m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/de66ceaa3da540898ca3c5ff8e313f52 : SUCCESS in 7m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d0527af4ecf5481292b6b41b00025df5 : SUCCESS in 39m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/02871f46198448c6a04058833a60a866 : FAILURE in 6m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4c33df83f99b451496f7675caf2c8c80 : SUCCESS in 1h 12m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/98f5db7b08164bb99add8a40ec5fc76a : SUCCESS in 1h 25m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2f12e705790d4f80beb076242aea6691 : FAILURE in 1h 30m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4090d989a6f44df19d234b44a5ec5c07 : SUCCESS in 1h 12m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0dad0ea70ab749d8af2c748baba886e1 : SUCCESS in 1h 19m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/35af9e0872d04056b245ed28967c6e05 : SUCCESS in 40m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bf4f04a8e4e74a969702d826fca79191 : SUCCESS in 1h 15m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b3eec5a689bb42e6997686e875727a03 : SUCCESS in 2h 11m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebf4de63b02f47388fb1b8c8a9aa7d62 : SUCCESS in 53m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7355f8881aef4ecdb20fcfd60a0ac28c : SUCCESS in 46m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d128f68b824c4a21cec1ca0720fba5cffdf8da9b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-04 16:29:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/5710965fa18047c0a97693b391732ffb : RETRY_LIMIT in 2h 07m 46s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/95/816695/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3873b63/ : SUCCESS in 1h 00m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f0efc4c621a3dc9fab4a87d8092dbe21c91af33c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-04 16:35:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/816695/1/check/dsvm-tempest-cisco-zonemanager-job/967adc5/ : SUCCESS in 1h 07m 55s","accounts_in_message":[],"_revision_number":1},{"id":"03895d60cad826358db13f051d9916d950dd44b9","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-11-04 16:40:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09917735998c04576cdc4497827f5f8b4b2bb75a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-04 16:55:16.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/816695/1/iscsi/ : FAILURE in 13m 49s\n- cinder-smb http://cloudbase-ci.com/cinder/816695/1/smb/ : FAILURE in 13m 55s","accounts_in_message":[],"_revision_number":1},{"id":"bdb8072fb224ba00b9fd552c23299ccdff86d5a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 17:10:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d935bef909df49699a2e8cc32d700b12 : SUCCESS in 18m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4878efefbf2a4ba181020d15029f9869 : SUCCESS in 23m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"efc3046f4795a5e505cdfe7d7db281b6f43a2b7e","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-11-04 17:44:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/6d234ee/index.html : SUCCESS in 1h 01m 07s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/cd1969b/index.html : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":1},{"id":"aee048d0cd96cb3ac8694a6cd42b3c1368d301aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 18:58:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28ade7ce815b4ed2969eab703cf14417 : SUCCESS in 9m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/912e6fd481834171a8afba6da04bb58e : SUCCESS in 12m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/141418d9f8f64b7a9d63e1758c2dc5d5 : SUCCESS in 11m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fefbaf82d24446e29656e2b3174e75b1 : SUCCESS in 13m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b0e8642c31f46768c21f9f4bbb18a2e : SUCCESS in 16m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/4f463ea5f04a40199b848e4625c10a00 : SUCCESS in 59m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/665aad1c11794ab18ea282b0b3c45808 : SUCCESS in 1h 19m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2353b05478c245afb1087f8aad0ecd7a : SUCCESS in 8m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e0e1d38796ad4f1fa4c08045bc9904db : SUCCESS in 16m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/46215dc2860e42deb2c0f9f60029b3a2 : SUCCESS in 8m 55s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7b97013943034bc8a7743b801f8ed451 : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/55a2bea4511541f48bc8bf1bc4cc6009 : SUCCESS in 8m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/74fa21237e734565a96c9c5f50c3773c : SUCCESS in 7m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/588d3c6cbb7b474ea3d9106bc1c39a60 : SUCCESS in 47m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f548555a15ca4002b110a73d2da87e12 : FAILURE in 7m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/91add97557a94737a3d682f557c4e3bd : SUCCESS in 1h 08m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/da614213ab1f4019806c431d45851cdb : SUCCESS in 1h 25m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fbbe7a3bd08f470a83cf438d34eb926d : SUCCESS in 1h 31m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ee41c57520a14a1fb06e9eb4ad961d37 : SUCCESS in 1h 09m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ad7422eb8b9e41c8909ccbafd3c955b9 : SUCCESS in 1h 14m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/734b4fe8955545af989ab6e930e101d8 : FAILURE in 53m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e7ffa287c2cb4a549255c3c87385d48e : SUCCESS in 1h 22m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/795499014749424fbcdd09f4ef7d0275 : SUCCESS in 1h 52m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a701527fd7dd497e90c8418a676b0287 : SUCCESS in 1h 01m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f662fedc885345169735ec61468d2470 : SUCCESS in 56m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b20e76b45913425af896d26f26fd42a6ab16c416","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-04 19:08:39.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-816695-1/ns5_nfs : SUCCESS in 4h 19m 3s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-95-816695-1/ns5_iscsi : SUCCESS in 4h 10m 45s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"ea5ee642820de93fcb558367c5dfd27b88818602","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-04 19:33:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/95/816695/1/upstream-check/cinder-cDOT-iSCSI/f021235/ : SUCCESS in 32m 59s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/95/816695/1/upstream-check/cinder-cDOT-NFS/2c921c1/ : FAILURE in 7m 01s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/95/816695/1/upstream-check/cinder-cDOT-FCP/9e0b73a/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"87dfe137158761bf0105a0bc3a45a7bbb42523fb","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-04 20:26:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/95/816695/1/20211105023316/NEC-FC : SUCCESS in 2h 51m 11s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/95/816695/1/20211105023316/NEC-ISCSI : SUCCESS in 2h 52m 47s","accounts_in_message":[],"_revision_number":1},{"id":"14562e77324fecf093a506e646436ad8c94c8e29","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-11-04 20:30:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/95/816695/1/upstream-solidfire-check/cinder-solidfire-iSCSI/c179ee5/ : FAILURE in 1h 30m 28s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"b3dbbcf4b48e2a2d3cbc3171e4bdfe3c14428d6b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-05 07:54:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/816695/1/check/dsvm-tempest-cisco-zonemanager-job/c64598c/ : SUCCESS in 1h 07m 41s","accounts_in_message":[],"_revision_number":1},{"id":"154893aec1da676aafe74f0c28a68b4e252ff020","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-11-09 12:59:48.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/95/816695/1/check/dsvm-tempest-TOYOU-ACS5000-driver/ab74361 : FAILURE in 9m 35s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/95/816695/1/check/fc-test/b08c8f5 : FAILURE in 4h 30m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ecd1609b664b4571f774b9618ffc3216d38aa641","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-11-12 13:17:22.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a53a497ba739f624f04792fab3b93e58bff05867","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-11-18 09:26:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ee8fbf50712fb9b69c734e36982a2f3cdc880a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 09:27:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2058d54dbaac12dd8e263f752dcc603ac477c86e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 11:15:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ef20940b912458b954563f1a8c9e792 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ac8d8d39bb74898a56e3df853ff95c5 : SUCCESS in 9m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bde0998b4df4baf972c2526daa06281 : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a610290bdc54d2594d8b37f22c2cf0c : SUCCESS in 13m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/039f35f25cae4c8999f0ed33d957d975 : FAILURE in 1h 09m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/853035c142dd406cb5c4b9f65947bef3 : SUCCESS in 1h 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce55e63b99334a52a5bfa1ab23d5e58d : SUCCESS in 9m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/030b659afb7548a2a63576e8ffe68e91 : SUCCESS in 1h 18m 34s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/971c28e44c4f48c2965101d5ab33caa7 : SUCCESS in 1h 38m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cafa6bf784b843b88a05305e047532cc : SUCCESS in 44m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d04283320df948c885e6ae1b6f2672f2 : SUCCESS in 46m 38s","accounts_in_message":[],"_revision_number":1},{"id":"54d87024d821c12924180ded22dc90ec92d9766f","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-11-18 17:32:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e556beb2f6e53d822903360d8943cc89fe6b608a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 17:55:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dff91ac12c9e40f88c56ab7483774502 : SUCCESS in 20m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a65c76697e8a4b7b9feaf5908d4452fe : SUCCESS in 16m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e7a3ba0dcbb5a66a087dfffc6fcf0304a82e5464","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-18 18:33:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/816695/1/iscsi/ : SUCCESS in 1h 00m 42s\n- cinder-smb http://cloudbase-ci.com/cinder/816695/1/smb/ : SUCCESS in 50m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0cc7892af6a5e3f99abaa759389b0e35e01a0367","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-11-18 18:38:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/4615f5b/index.html : SUCCESS in 1h 05m 19s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/e740071/index.html : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":1},{"id":"9de9ba3dad50929a2b602246df1e1182be67e05f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-18 19:36:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/95/816695/1/check/dsvm-tempest-cisco-zonemanager-job/5f2fbdf/ : SUCCESS in 1h 06m 59s","accounts_in_message":[],"_revision_number":1},{"id":"95d330b907792c82c48e87b57886997374a44221","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 19:48:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa0dc71dac13485b8fea592c372450b5 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66d3f1a381bc4db995dd52a1b2116a03 : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33571d992f5a47a9982fe29032345a9a : SUCCESS in 12m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f26d13c3b0a141adb4be88504d0993d7 : SUCCESS in 11m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92e753b363f941df9f2149425ea58301 : SUCCESS in 12m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/042abb2692984ce689a48358edef29bf : SUCCESS in 1h 06m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b1cdf514abb741cb86ecf35c9b71b3e7 : SUCCESS in 1h 21m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b89841bba18f441387a4e00cad1ec9fa : SUCCESS in 7m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2d36da1ee21240f497b9782158b62f02 : SUCCESS in 15m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/735cd0d578fe45128a21edf8aa4a01ca : SUCCESS in 6m 15s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2e241777dab34af38b3d23f9a7594305 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7e82d49f266f4aa880071fa5867f127f : SUCCESS in 6m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/abe3e89d6e3d4ba49539a51cc16eadfb : SUCCESS in 7m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/044535af659b46d2ab5ee47dfe1e2891 : SUCCESS in 32m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d6b3fc6a4df84aac96ea92a8cfb4cf62 : FAILURE in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/394691f0f26f45b7b0c09035c11f73e8 : SUCCESS in 1h 16m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/87c1dea338a24689be728b32dd9d7138 : SUCCESS in 1h 33m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6f006d0be69f4f5dbc54400fbf5d48af : SUCCESS in 1h 47m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/efa4f77518984e29bfc4fe172f12fec1 : SUCCESS in 1h 23m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0f3e1cb0e4994733a8a4cd02270944f2 : SUCCESS in 1h 23m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8991e94afa4d4a86aa9cbe972cae9fe1 : SUCCESS in 40m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0e315ba6e36946f7956508ef83453b6e : SUCCESS in 1h 38m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/671dba8b565d4dc3bf5160bf601d651e : SUCCESS in 2h 08m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb8e032062bc4244a8e0192e3715333e : SUCCESS in 52m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39b58cfe5dfa48319ba2f682482012cd : SUCCESS in 45m 59s","accounts_in_message":[],"_revision_number":1},{"id":"91dccd8d1dd387603e982fe0d335be02450c8f2f","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-11-18 19:50:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/95/816695/1/upstream-solidfire-check/cinder-solidfire-iSCSI/53cf2c2/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"58f4f370062b0e849ac24c217aa4ad1424b9fa79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 19:50:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3fb033610a6ad3875f22ec9e22e483448c8f956","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-18 21:06:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/95/816695/1/upstream-check/cinder-cDOT-iSCSI/ac528b3/ : NOT_REGISTERED\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/95/816695/1/upstream-check/cinder-cDOT-NFS/8cf4cb9/ : NOT_REGISTERED\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/95/816695/1/upstream-check/cinder-cDOT-FCP/6d208dc/ : SUCCESS in 1h 16m 19s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"022383a63c580800d3b0c0ef682807f24017e6c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 21:12:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c8c352f3afd4fca99a3e869697b0c1e : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64f22d78ca7c40ee8b6d6aac90bc1eee : SUCCESS in 12m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6645bde36aa45789d45d43ee32ca95f : SUCCESS in 11m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c702709ef6874d389792e99e9e9ab37b : SUCCESS in 11m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/3940b3182b754fd399407b0b6820765e : SUCCESS in 1h 02m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4ab6948e477740829064f3470a2d282c : SUCCESS in 1h 12m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/feec68b410264b6bb089fd1ba89725ec : SUCCESS in 9m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9cf26adc946f4daab1570dd13eeb6ba2 : SUCCESS in 1h 17m 38s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/669c92630d384e49beb4729a0d5f3051 : SUCCESS in 1h 18m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f5ee728eb9b49f8923a81428a3dd01f : SUCCESS in 52m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8cc262debbf5436ca7957e4c46cd8ea8 : SUCCESS in 48m 16s","accounts_in_message":[],"_revision_number":1},{"id":"1f1678fd1813b8439fabab2a6f3ab3a56a5bd11c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 21:12:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cbd302d7a4cfc6e74620d0e21f96fd485c27d20e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 21:15:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fee8b65e8294139869b412cc2617f24 : SUCCESS in 1m 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a90cd3df8cd4eec8460f3692e540a9a : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe1e6be52f84d1aafdb4a140dfdaf4d0277b9623","revisions":{"fe1e6be52f84d1aafdb4a140dfdaf4d0277b9623":{"kind":"REWORK","_number":1,"created":"2021-11-04 13:30:51.000000000","uploader":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"ref":"refs/changes/95/816695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/95/816695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/816695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/816695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/95/816695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/95/816695/1"}}},"commit":{"parents":[{"commit":"73fb00bb691f0e7c84f08785969e9ed093d7817e","subject":"Merge \"PowerMax Docs - Xena release notes\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/73fb00bb691f0e7c84f08785969e9ed093d7817e"}]}],"author":{"name":"Helen Walsh","email":"helen.walsh@emc.com","date":"2021-03-05 12:56:28.000000000","tz":0},"committer":{"name":"Helen Walsh","email":"helen.walsh@emc.com","date":"2021-11-04 13:30:49.000000000","tz":0},"subject":"PowerMax Driver - Fix for legacy PowerMax OS around generations","message":"PowerMax Driver - Fix for legacy PowerMax OS around generations\n\nIn the previous version of PowerMax OS generations of snapVx were\nused instead of unique snap ids which were just introduced.\nA generation can be returned as a 0 integer which equates to False\nin python.  The fix is to convert the int to a string if it is\nreturned from REST as an int.\n\nCloses-Bug: #1938572\nChange-Id: I5b660776190f3026296d6d3237bd3b0d609f769f\n(cherry picked from commit ee1b5e2b756f49a384d348555e9611526c0c1217)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe1e6be52f84d1aafdb4a140dfdaf4d0277b9623"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe1e6be52f84d1aafdb4a140dfdaf4d0277b9623"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
