)]}'
{"id":"openstack%2Fcinder~817135","triplet_id":"openstack%2Fcinder~master~I0c6cd0bc7741656bee004c3e3c74f772e5691202","project":"openstack/cinder","branch":"master","topic":"xtremio_ig_conflict","attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2021-11-24 04:21:38.000000000","reason":"Keigo Noha replied on the change"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2025-11-10 08:06:46.000000000","reason":"\u003cGERRIT_ACCOUNT_25613\u003e replied on the change","reason_account":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"}},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2022-04-06 15:51:50.000000000","reason":"Someone else replied on the change"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2021-12-08 06:37:11.000000000","reason":"Keigo Noha replied on the change"}},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-06-17 16:23:40.000000000","reason":"Brian Rosmaita replied on the change"},"31016":{"account":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"last_update":"2021-11-28 21:40:14.000000000","reason":"Ivan Pchelintsev replied on the change"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2022-03-29 15:13:25.000000000","reason":"Eric Harney replied on the change"},"25613":{"account":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"last_update":"2025-11-10 08:06:46.000000000","reason":"\u003cGERRIT_ACCOUNT_25613\u003e replied on the change","reason_account":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"}}},"hashtags":[],"change_id":"I0c6cd0bc7741656bee004c3e3c74f772e5691202","subject":"Retry _get_ig() when Initiator Group already exists.","status":"NEW","created":"2021-11-09 03:44:51.000000000","updated":"2025-11-10 08:06:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":32,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"724a1095e27b4564a9fdbdb8f9cc9611158c3f2e","_number":817135,"virtual_id_number":817135,"owner":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-06 16:07:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":1,"date":"2022-06-17 16:24:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":-2,"date":"2025-11-07 15:45:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2022-04-06 13:36:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2022-07-11 10:35:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-31 15:44:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-09 04:03:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-09 04:18:03.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-09 04:50:09.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-09 05:31:11.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-09 05:34:44.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-11-09 05:35:57.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-09 05:41:16.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-09 05:44:25.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-11-09 06:36:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-09 06:52:49.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-09 07:48:09.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-11-09 07:48:58.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-09 13:27:20.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-10 07:11:28.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2021-11-10 08:26:39.000000000","updated_by":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"reviewer":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"state":"CC"},{"updated":"2021-11-10 08:51:31.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-11-10 09:06:52.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-10 09:25:58.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-11-10 09:36:15.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-11-10 09:45:23.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-11-10 09:58:25.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-11-10 10:00:25.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-11-11 00:47:00.000000000","updated_by":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"reviewer":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"state":"REVIEWER"},{"updated":"2021-11-16 10:07:54.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-11-18 08:34:23.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-11-24 04:21:38.000000000","updated_by":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-11-24 04:21:38.000000000","updated_by":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"reviewer":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"state":"REVIEWER"},{"updated":"2021-12-08 06:37:11.000000000","updated_by":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-12-08 06:37:11.000000000","updated_by":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-03-16 18:41:39.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2022-03-18 03:11:45.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-03-18 04:10:59.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-30 02:44:51.000000000","updated_by":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"reviewer":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"state":"CC"},{"updated":"2022-03-30 04:18:22.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-30 05:02:50.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-03-30 12:26:52.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-30 15:43:47.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-03-31 12:03:16.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2022-04-06 13:36:47.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2025-10-31 15:44:55.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2025-11-07 15:45:52.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"580de1797c507deb39443afe5158d8aea54ae05b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2021-11-09 03:44:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd5c2ee3a34f81bbb54d6766f8a7463ee5d308c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 04:03:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9b614c7e4fa4713a507a12de563a0ee : FAILURE in 15m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/757034fdd5c540dfb71765ef62b21ea3 : FAILURE in 15m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"376cb9244c33e967e3ac3ddff2b2eb2a2841dc35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2021-11-09 04:16:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"942040a046640a34c000ab750583280015adc5e6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-09 04:18:03.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/35/817135/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"6d6869f636c086cf0b16f0867c936cc60d863992","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 04:50:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/62302dd2707b4e3b89f30550fc08ec2f : SUCCESS in 20m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e7577f0b02cf4a67a5c0d40b1e0c74fb : SUCCESS in 26m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa06b2aa7742085ead354300e21af40437f31d7d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-11-09 04:50:09.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/817135/817135/2/check-not-vote/ext-cinder-zuul/d2c45e0 : SUCCESS in 31m 47s","accounts_in_message":[],"_revision_number":2},{"id":"46d1d8d8f82cf61adcb3a02a613d62e2fedda654","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-09 05:28:38.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/35/817135/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"d96e00f00be23ba34ce7ebc0f4ceed163fbb0bf3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-09 05:31:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/35/817135/2/check/iscsi-driver-nimble/a7e155d/a7e155d.html : FAILURE in 1h 09m 52s","accounts_in_message":[],"_revision_number":2},{"id":"71ebcd87ed9cdf24cf60a1dd532bd3dd596c7763","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-09 05:34:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/35/817135/2/check/dsvm-tempest-cisco-zonemanager-job/405aece/ : SUCCESS in 1h 11m 02s","accounts_in_message":[],"_revision_number":2},{"id":"b61068e19b26ff6973f13348338490e4d8eb1d06","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-09 05:35:57.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/817135/2/check/devstack-plugin-ceph-dsvm-focal-py3/ec42711/ : FAILURE in 57m 32s","accounts_in_message":[],"_revision_number":2},{"id":"2ac1ec1fa753b4332ba6f453a4bc99342ee9c4d0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-09 05:36:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/35/817135/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/77ece3a/77ece3a.html : SUCCESS in 1h 17m 21s","accounts_in_message":[],"_revision_number":2},{"id":"120a3645179fc816af541e1de3713f231bc7cbdf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-09 05:36:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/35/817135/2/check/iscsi-driver-primera/cca5825/cca5825.html : SUCCESS in 1h 17m 57s","accounts_in_message":[],"_revision_number":2},{"id":"fbcedc42884fc843b06fd8885c6fadbc47380198","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-11-09 05:41:16.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/15d0f814803542f1bfa4b2e624e68ebe : SUCCESS in 1h 21m 25s","accounts_in_message":[],"_revision_number":2},{"id":"5698a89228226afcee7c429cd9bfbb6f7618e16c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-09 05:44:25.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817135/2/iscsi/ : FAILURE in 1h 26m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/817135/2/smb/ : SUCCESS in 1h 01m 58s","accounts_in_message":[],"_revision_number":2},{"id":"6d571eb589e3d64eaa510a80e9a60c1171824673","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-09 05:59:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/35/817135/2/check/iscsi-driver-alletra-6k/4e1a68c/4e1a68c.html : SUCCESS in 1h 40m 59s","accounts_in_message":[],"_revision_number":2},{"id":"f6e629bccb50729a6b2817681232bcbab127612d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 06:36:30.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d7cd8d7e3db4a039e7be8a9ac3447c2 : FAILURE in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c9014dfc93b4cc58ca10a2bf310e4a4 : SUCCESS in 9m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f0130de1cbd469780b2722db3e28cfa : SUCCESS in 10m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09712002375940a5b8df1e3f13317524 : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89134cde17334717a539a5a1fb9d92be : SUCCESS in 11m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/d7fc0e9309214aefb6a801d2eb686df3 : SUCCESS in 1h 04m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7596c1557e3a46a687e2c517ba852ac1 : SUCCESS in 1h 14m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2e258a2a65c34e8c9f28296f7c8fc285 : SUCCESS in 13m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e9f551ead4684a6cab7e78a8f400b866 : SUCCESS in 5m 45s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e22401d90cf4108b97a520c61c52ff7 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2ed41d712acc41e1a08d5487b8d23167 : SUCCESS in 6m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e5f01cb092b84875b4ca73e5a936cd8f : SUCCESS in 7m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b84e2ffc8ddc47f49b7b43994bb94eab : FAILURE in 33m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9dcfd0ae1a444fadaf684421a47210da : FAILURE in 4m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/03db99021e9e463699b744ab9600d052 : SUCCESS in 1h 24m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/10b868e95252470b88ecb6b55e830119 : SUCCESS in 1h 21m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7325ee27549d442d8cd02a1e431cbc96 : SUCCESS in 1h 37m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e6e1d8f8d8354c5aac90f12b1ea3d01f : SUCCESS in 1h 15m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5cc5a028b9e14865a3d9909964ea211e : SUCCESS in 1h 43m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/de6edae498704a11a59344cb70dea028 : SUCCESS in 48m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ae2edc81891b4d0f9e77ed3208543e9b : SUCCESS in 1h 21m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/644df243470047989da3a0f9cb427f15 : SUCCESS in 2h 13m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02cfe6c96c354eb59b09efdfd6394f27 : SUCCESS in 50m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/467d1f69201645ed86530f58b5bb3ec2 : SUCCESS in 43m 48s","accounts_in_message":[],"_revision_number":2},{"id":"1607d6de67162eeaf53ee6e6c2eb4d03e6244223","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-09 06:52:49.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/35/817135/2/check/DellEMC_XtremIO_FC/c658052/DellEMC_XtremIO_FC/89 : SUCCESS in 1h 11m 53s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/35/817135/2/check/DellEMC_XtremIO_iSCSI/89de495/DellEMC_XtremIO_iSCSI/116 : SUCCESS in 1h 54m 08s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"3b8f047a89ff0fecf2a6f4114326937098ecda7e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-09 06:52:51.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/35/817135/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"aa3cb744d094208a5f016507528ce4a26f36cddb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-09 07:48:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/35/817135/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/865d291/ : SUCCESS in 2h 49m 55s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/35/817135/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/900276c/ : FAILURE in 2h 24m 10s","accounts_in_message":[],"_revision_number":2},{"id":"98f8fcafc0b9ea5aacfd664fbffde32ad44d5dce","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-09 07:48:58.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-35-817135-1/ns5_nfs : SUCCESS in 3h 56m 9s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-35-817135-1/ns5_iscsi : SUCCESS in 4h 3m 27s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"7cfc579114148ab886d840e8430bff1ebfc9e0cb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-09 07:49:58.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/35/817135/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"c59ef190db9768972757f9e1a685261b783961c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2021-11-09 08:07:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b24c1af9d7287634c9873cfe5479dbe7f4dd002e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-09 08:07:38.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/35/817135/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"df746facfe5d1f846bdd3a2ddb905b3d1f5c94e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 08:35:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/803ca294b19f4c41909d517c051de9e5 : SUCCESS in 19m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b334d5e221a45cebc347e690da171dd : SUCCESS in 24m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1d686d5647e672507f10e0f60e72ea212c12042e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-09 08:37:33.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-35-817135-2/ns5_nfs : SUCCESS in 4h 1m 39s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-35-817135-2/ns5_iscsi : SUCCESS in 4h 7m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"fff99095af6ec1faa5c30e91fecdafaf91a4d9e6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-09 08:38:23.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/35/817135/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"6c3711e658e82fa4a30ee3f645b541d0568e144b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-11-09 08:44:41.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/817135/817135/3/check-not-vote/ext-cinder-zuul/d6b282d : SUCCESS in 35m 12s","accounts_in_message":[],"_revision_number":3},{"id":"6551b8b4e7abad3b389453d4ccb4ce68057e604d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-09 09:20:43.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/817135/3/check/tempest-dsvm-full-focal-py3/918fcb2/ : FAILURE in 53m 43s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/817135/3/check/devstack-plugin-ceph-dsvm-focal-py3/8f2bb84/ : FAILURE in 1h 05m 27s","accounts_in_message":[],"_revision_number":3},{"id":"0ed4a8fa113b78f6ab55376be9f41f4fa67df60d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-09 09:21:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/35/817135/3/check/dsvm-tempest-cisco-zonemanager-job/a676144/ : SUCCESS in 1h 08m 47s","accounts_in_message":[],"_revision_number":3},{"id":"c93b212bd4f9d5db892994fe769ffbbcbb1144e9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-09 09:21:40.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/35/817135/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"e00d5d402045d44208a1fe3fba2ba93b135eac10","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-09 09:22:32.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817135/3/iscsi/ : SUCCESS in 1h 11m 46s\n- cinder-smb http://cloudbase-ci.com/cinder/817135/3/smb/ : SUCCESS in 1h 14m 51s","accounts_in_message":[],"_revision_number":3},{"id":"e1f701fdd039786f2ad18d103525b8d5273dcc66","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-09 09:25:45.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/35/817135/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/69edeb5/69edeb5.html : FAILURE in 1h 16m 29s","accounts_in_message":[],"_revision_number":3},{"id":"b86eedde890ff4b1a372f7f30e91ec9f90060cf7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-09 09:28:39.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/35/817135/3/check/iscsi-driver-primera/6a4ab5d/6a4ab5d.html : FAILURE in 1h 19m 03s","accounts_in_message":[],"_revision_number":3},{"id":"ea15e82f446d349d1222c2b55088d209046e1925","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-11-09 09:33:38.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/59133d537fd645769bf4cca1ee75e3d4 : SUCCESS in 1h 23m 58s","accounts_in_message":[],"_revision_number":3},{"id":"e3445e869e216df49e2c01eb04f024ff260b6533","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-09 09:41:43.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/35/817135/3/check/DellEMC_XtremIO_FC/8762c21/DellEMC_XtremIO_FC/90 : SUCCESS in 1h 01m 46s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/35/817135/3/check/DellEMC_XtremIO_iSCSI/49ce1cc/DellEMC_XtremIO_iSCSI/117 : SUCCESS in 1h 34m 23s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"ab1f9bead03c957fd815d44587f336fbc0fedcfc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-09 09:42:11.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/35/817135/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"ea408dbd55d44de918edcd3a2c07cdf3787bfe1e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-09 09:49:20.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/35/817135/3/check/iscsi-driver-alletra-6k/fed38b4/fed38b4.html : SUCCESS in 1h 40m 23s","accounts_in_message":[],"_revision_number":3},{"id":"b0ea8c563793187cdf07e5a089c92be61d083c60","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-09 09:59:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/35/817135/3/check/iscsi-driver-nimble/39b6950/39b6950.html : SUCCESS in 1h 50m 03s","accounts_in_message":[],"_revision_number":3},{"id":"c841b37e2e729d549dc0a3e3a9e2b0cf0ac5ddfb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-09 10:36:27.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/35/817135/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5f80e11/ : SUCCESS in 1h 32m 37s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/35/817135/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/8719ae8/ : FAILURE in 2h 28m 23s","accounts_in_message":[],"_revision_number":3},{"id":"6e7c5a9b0d6bd98b90bedab379748b08138aceb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 10:42:07.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c316e6610c9a4d058418c6d13fce7fe3 : FAILURE in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ef5100d2b934d80b36de88b62621015 : SUCCESS in 11m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03ae954ee9db481fae2e898eca053eb6 : SUCCESS in 12m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22988b0b625d47c193b64c055253a4f6 : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f4155598dd467da9227d2bf720fb03 : SUCCESS in 12m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/0ca730949c49495f9edb8168eee89b3b : SUCCESS in 52m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2aaf837450964057926cda20359c3abc : SUCCESS in 1h 13m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/386346fab72f4119b0db716f54264908 : SUCCESS in 14m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2d7261aa17704fdf8bfe678708574c23 : SUCCESS in 7m 17s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8b821e92297a44c9a19b8dd2f796b5ed : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7f554119722b47db813bc9af83094d1a : SUCCESS in 8m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/27b097a5ef3640c8a601c77abfe5da46 : SUCCESS in 6m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ecfed7d10de6446a88a2a618e803d288 : FAILURE in 32m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8c18533cba5544e594c1228427850e3c : FAILURE in 4m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/85d1bcc41eeb40b7965107e31dc4af29 : SUCCESS in 1h 30m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5e7f2b8bdf514c2593552120b48607c3 : SUCCESS in 1h 21m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/50ffd476f4cd43baa9e788bae24e3186 : SUCCESS in 1h 31m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e108c0f0f43246b181e950ba3b139d2a : SUCCESS in 1h 20m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d3e4a3ffc0f044d9a0a8182fa5e79544 : SUCCESS in 1h 14m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bb8ac1ba3a084a4cae0f684af06a03c6 : SUCCESS in 40m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cb8c2b6792e046c9bb309db83e075002 : FAILURE in 13m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/af0533921ac84ea8b8806dd89678fd3b : SUCCESS in 2h 31m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0a96b1bee2d4d56be7f5f399d791379 : SUCCESS in 53m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4801bf5940c8483982ba66101d76a66f : SUCCESS in 45m 47s","accounts_in_message":[],"_revision_number":3},{"id":"3786aa7bdca7762b4e59c9c060692a28e9487b65","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-09 10:48:54.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/35/817135/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"1f46c5a4d7cda5206b84375490117b51e8fb78c5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-09 12:10:30.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-35-817135-3/ns5_nfs : SUCCESS in 3h 57m 18s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-35-817135-3/ns5_iscsi : SUCCESS in 4h 2m 52s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"df1235e54fd25895205f7e5aa0b57df62e032bd3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-09 13:26:17.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/35/817135/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"76a7dd17b76b2fac25d329462ec31b463df43376","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-09 13:27:20.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/35/817135/3/20211109192640/NEC-FC : SUCCESS in 2h 52m 55s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/35/817135/3/20211109192640/NEC-ISCSI : SUCCESS in 3h 00m 36s","accounts_in_message":[],"_revision_number":3},{"id":"aeb2a99efd38dd08c6755c158aa2c6504035ef26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2021-11-10 04:24:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"40530c438afe1933c48b6061d992e11b0d7f0d87","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-11-10 05:00:13.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/817135/817135/4/check-not-vote/ext-cinder-zuul/0f3b32d : SUCCESS in 34m 33s","accounts_in_message":[],"_revision_number":4},{"id":"b019e19a7fc2646c5d875f6b345bc0e18534217b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 05:06:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f92028baf39647efa5ff07e5a4dd220b : SUCCESS in 17m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43a098f0c6554476875a4082b8e663c5 : FAILURE in 30m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7dfbef91c544f4cca183bd328b75408aba56348b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-10 05:32:14.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817135/4/iscsi/ : SUCCESS in 57m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/817135/4/smb/ : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":4},{"id":"28ffb59f381585a6c1cae9266d1ef8d2f5b4ebdf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-10 05:38:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/35/817135/4/check/dsvm-tempest-cisco-zonemanager-job/92edd8a/ : SUCCESS in 1h 08m 43s","accounts_in_message":[],"_revision_number":4},{"id":"9d23a4c53103fe6e1ed94db50263ea42c94b19cc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-10 05:38:59.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/35/817135/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"d507ef8a8433bd64d3803dbab635e7659f2b94a3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-10 05:42:54.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/35/817135/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/cef1762/cef1762.html : FAILURE in 1h 16m 32s","accounts_in_message":[],"_revision_number":4},{"id":"d81780c973bcd9544b72f6c7d8e3652ea44b6b50","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-10 05:45:24.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/35/817135/4/check/iscsi-driver-primera/e72e27f/e72e27f.html : SUCCESS in 1h 18m 47s","accounts_in_message":[],"_revision_number":4},{"id":"726e70aabf073aecbbfddd777a7aefe535d51143","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-11-10 05:57:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4c04b4f9c7604b09845c62704b85e90d : SUCCESS in 1h 30m 44s","accounts_in_message":[],"_revision_number":4},{"id":"1de340a3a728e084c9c7132b3be378a6551574ba","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-10 06:00:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/35/817135/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4b0f857/ : SUCCESS in 1h 34m 42s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/35/817135/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b1d51a6/ : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":4},{"id":"6742b8afcbff333e2f669cd81300ae5330d7c0b5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-10 06:06:57.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/35/817135/4/check/iscsi-driver-alletra-6k/3d783e3/3d783e3.html : SUCCESS in 1h 40m 59s","accounts_in_message":[],"_revision_number":4},{"id":"1fca0ca244c384d919041ac190e7c44e32203fd5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-10 06:15:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/35/817135/4/check/iscsi-driver-nimble/92ea03e/92ea03e.html : SUCCESS in 1h 49m 13s","accounts_in_message":[],"_revision_number":4},{"id":"7cb62e95a891106984456e9998d24b877ad6105f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-10 06:18:29.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/35/817135/4/check/DellEMC_XtremIO_FC/5cbff06/DellEMC_XtremIO_FC/95 : SUCCESS in 1h 02m 21s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/35/817135/4/check/DellEMC_XtremIO_iSCSI/90986ba/DellEMC_XtremIO_iSCSI/122 : SUCCESS in 1h 53m 38s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"e3a60b85ea4d2d90e609d3d963faf649608ff84a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 06:54:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3212bc3ba4cd41efb19a5585fd702b96 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8cd842ca2ad497aabd8681f1261e723 : SUCCESS in 11m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc5e366560f34186907f255aff6e384b : SUCCESS in 9m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf6c86817f754b519739b3c801554e5a : SUCCESS in 11m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eab2fc5916ac4d89a28f8a113b008f85 : SUCCESS in 11m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/a4ff67a215474e6ab02eeed00e64ca34 : SUCCESS in 1h 05m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a69ba68499e4475a99dff2703445a025 : SUCCESS in 1h 14m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/58255cf5dcac404a8a3a1456bf7b271c : SUCCESS in 14m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7692783f439b423599cabcdf5b414353 : SUCCESS in 6m 22s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/34168340a8be4410a6b71da41b86c759 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/335c96f1def24b4a9aa4a7b511312fed : SUCCESS in 6m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/200badafd50241e0b9d12e8654f21d3e : SUCCESS in 6m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a62f48e18f3a4e649f64370b92c6e640 : FAILURE in 36m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/237c052ef82649a0bf940b79051fb45a : FAILURE in 4m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/150d81ada58749fdb8bca89a050675dc : SUCCESS in 1h 02m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1644516b42694513bc4a9c54beb8f76e : SUCCESS in 1h 39m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/98381f2fc97e40f3babd48f9ada0a51c : SUCCESS in 1h 37m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e5f25ea659994bc09de14365c3efffbd : SUCCESS in 1h 06m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7ea44cfbb07f48db87ef2b7ac166f58b : SUCCESS in 1h 30m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/aaf4cda283d84132a15b160d92efd9ab : SUCCESS in 48m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1f2832e7678a4ea1961cf650271671c2 : SUCCESS in 1h 22m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/70acbe22bc924cf48635bdda26a23ef7 : SUCCESS in 2h 21m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/296fdba4341c4de99c487fd54f873ac5 : SUCCESS in 55m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48e73d55fa864e87bdfbebb104d182ff : SUCCESS in 52m 51s","accounts_in_message":[],"_revision_number":4},{"id":"1cb906a37abb5da330468510d525e68e43929cb4","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-11-10 07:11:28.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-35-817135-4/1636527260 : SUCCESS in 17m 07s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"28786954edf0720a9b9e7ad125fd9ddf61b7b4ac","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-10 07:37:44.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/817135/4/check/tempest-dsvm-full-focal-py3/fdf3151/ : SUCCESS in 1h 55m 57s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/35/817135/4/check/devstack-plugin-ceph-dsvm-focal-py3/708f7f4/ : TIMED_OUT in 3h 08m 27s","accounts_in_message":[],"_revision_number":4},{"id":"e24dc4d483555a3889ff411f808e8eb89ce2c614","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-10 07:37:48.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/35/817135/4/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"7be83273a0d55e5e80966309ade37d6edb5e3f28","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2021-11-10 08:26:39.000000000","message":"Patch Set 4:\n\nBuild Successful. ibm-gpfs-dsvm-tempest-full: https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/119/index.html","accounts_in_message":[],"_revision_number":4},{"id":"ffb5fb7a7038b32e0812296ae42de20a1da3da32","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-10 08:40:10.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/35/817135/4/check/DellEMC_XtremIO_FC/4ad59e3/DellEMC_XtremIO_FC/96 : SUCCESS in 1h 03m 28s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/35/817135/4/check/DellEMC_XtremIO_iSCSI/11db48a/DellEMC_XtremIO_iSCSI/123 : SUCCESS in 1h 45m 34s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"cceda99ef816998fd98bb32e1b20b89fdc9f9ad5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-10 08:41:06.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/35/817135/4/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"b43e8eed1ea8a7b299c076f9af0a554e6ce6a444","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-10 08:51:29.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-35-817135-4/ns5_nfs : SUCCESS in 4h 4m 18s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-35-817135-4/ns5_iscsi : SUCCESS in 4h 8m 23s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"5216496ecc0864e0983e4fcca31007808fcee26e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-11-10 08:51:31.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/35/817135/4/check/DellEMC_PowerStore_iSCSI/55e3635/DellEMC_PowerStore_iSCSI/48 : SUCCESS in 1h 56m 53s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/35/817135/4/check/DellEMC_PowerStore_FC/e858f3b/DellEMC_PowerStore_FC/55 : SUCCESS in 1h 15m 20s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"1f352a093dc5450dbb957f54bb3a67ef0988a5f8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-10 09:06:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/35/817135/4/upstream-check/cinder-cDOT-iSCSI/c9f155b/ : SUCCESS in 2h 11m 38s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/35/817135/4/upstream-check/cinder-cDOT-NFS/eebf244/ : SUCCESS in 37m 16s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/35/817135/4/upstream-check/cinder-cDOT-FCP/1f1ccec/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"da7f544ce3baabf906b5f946198e8e95cd8f53a3","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-11-10 09:25:58.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- PVME-iSCSI http://os-logs.tristero.net/dell-ci/817135-4/PVME-iSCSI : SUCCESS in 1:01:16\n- PVME-FC http://os-logs.tristero.net/dell-ci/817135-4/PVME-FC : SUCCESS in 59:52.75","accounts_in_message":[],"_revision_number":4},{"id":"aa1d799a5dfcfe071b2e4559e00925ce8eef2dbe","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-11-10 09:36:15.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/817135-4/HPMSA-iSCSI : SUCCESS in 1:18:09\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/817135-4/HPMSA-FC : SUCCESS in 51:32.77","accounts_in_message":[],"_revision_number":4},{"id":"820c6f9d868d8b5548c65cc9d9eade2c61393e8f","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-11-10 09:45:23.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/817135-4/Seagate-iSCSI : SUCCESS in 1:31:16\n- Seagate-FC http://os-logs.tristero.net/seagate-ci/817135-4/Seagate-FC : SUCCESS in 49:51.89","accounts_in_message":[],"_revision_number":4},{"id":"46cbfbbc4c8d5c0d07fe61539ab30c75e8502e66","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-11-10 09:58:25.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/817135-4/Lenovo-iSCSI : SUCCESS in 1:34:21\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/817135-4/Lenovo-FC : SUCCESS in 59:32.47","accounts_in_message":[],"_revision_number":4},{"id":"74f748126e8ade5fe88b60c312e89b36dd7c4657","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-11-10 10:00:25.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/35/817135/4/upstream-solidfire-check/cinder-solidfire-iSCSI/557f3e4/ : SUCCESS in 3h 04m 56s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"7937ac6b95020356fb9316c5ab4160d23ac1cbe3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-10 11:37:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/35/817135/4/20211110174312/NEC-ISCSI : SUCCESS in 2h 52m 43s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/35/817135/4/20211110174312/NEC-FC : SUCCESS in 2h 54m 10s","accounts_in_message":[],"_revision_number":4},{"id":"cbe32eb3e358419ee45d165726b1f69782f30e29","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-10 11:50:24.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/35/817135/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"90382561f0fec1021c2ee0feea291cc164702ea8","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2021-11-11 00:47:00.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7997ac79a1af2aa5d4f24edb71d9ce4910fa7203","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-11-16 10:07:54.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/35/817135/4/check/dsvm-tempest-TOYOU-ACS5000-driver/7e0c454 : FAILURE in 12m 28s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/35/817135/4/check/fc-test/c878a00 : FAILURE in 1h 57m 52s","accounts_in_message":[],"_revision_number":4},{"id":"159a5766300fabdfa073fac369c07e96b47ed4dd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-16 11:14:57.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/35/817135/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"0316125298977311a6340eae33b53decf1d1bc6f","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-11-18 08:34:23.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"da8e7e7a8a417421d0b12809238596aa006ef1e5","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2021-11-24 04:21:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b1df20874f93636d617550d2ba58527dd5aec08d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-25 06:35:26.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-35-817135-4/ns5_nfs : SUCCESS in 4h 44m 43s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-35-817135-4/ns5_iscsi : SUCCESS in 4h 50m 28s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"98cc8f352b2672ebb8e7e3e7c5090177a82a0ba8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-25 08:21:45.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/35/817135/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"9ddfec3271e72f8db0ae7ceb53f716deb7565145","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2021-11-28 21:40:14.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8dc8a0a6a95b5d54ba0b4a117779e03184596d53","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2021-12-08 06:37:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"53a36f0c8fe5680cb4dfb9b183b028252a1b3469","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2021-12-22 01:20:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3f5efa6e3c71ff7f5303a62122f7cc98b1bd9df7","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2022-03-07 00:29:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f13dfd92684c529791751494e835e74e8410ac6e","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-03-16 18:41:39.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a68e4d5d1f0a83c2e0d75c3cfd73ca64e1ff2d23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2022-03-18 01:31:56.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"710582b899c20044c0e65f5332eef90857961298","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-18 01:32:06.000000000","message":"Patch Set 5:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15578/","accounts_in_message":[],"_revision_number":5},{"id":"702f206eab9aa9f6a7d34df4e3cf35270f3b9825","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2022-03-18 01:33:24.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e4f37ddb3d987248f1ef020ecaf40b87d2a2cde0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-18 01:33:41.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15578/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"4e15b3f7d02bff2034146362b98e3e780dbe1b9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 01:57:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/611ff999f3344778a20e7b56cc8ebaf3 : SUCCESS in 20m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bed8117f73f04c7c83e181fb5dd749ae : SUCCESS in 23m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3e66170bd2df39fd8b0c23cc3925c9f9ea03235a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-18 02:06:34.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/817135/817135/5/check-not-vote/ext-cinder-zuul/0dc874c : SUCCESS in 33m 50s","accounts_in_message":[],"_revision_number":5},{"id":"fc68d99e59c100ec7f575327e48e5bf732962ef5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-18 02:28:56.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/817135 : FAILURE in 55m 18s","accounts_in_message":[],"_revision_number":5},{"id":"d1372ecdfe3586347e4e6ea4634d1933579cf68d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-18 02:54:06.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/35/817135/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/181a9d4/181a9d4.html : SUCCESS in 1h 20m 25s","accounts_in_message":[],"_revision_number":5},{"id":"ddf1d7677548df8789da61175818b091b56cd4ac","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-03-18 03:01:44.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817135/5/iscsi/ : SUCCESS in 1h 24m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/817135/5/smb/ : SUCCESS in 1h 29m 12s","accounts_in_message":[],"_revision_number":5},{"id":"5c840c43ad3c38375dc2ac5345991abb6583cad4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-18 03:11:45.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/35/817135/5/check/iscsi-driver-alletra-6k/ba00b1f/ba00b1f.html : FAILURE in 1h 38m 40s","accounts_in_message":[],"_revision_number":5},{"id":"42e7f9f908634ab262156f2058406069066b3de7","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-03-18 04:10:59.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/35/817135/5/ : SUCCESS in 2h 10m 46s","accounts_in_message":[],"_revision_number":5},{"id":"c4d1ec17759152fc110ad59119c68f6e7d22f1b0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-03-18 04:19:06.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/35/817135/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/20e9763/ : FAILURE in 2h 45m 26s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/35/817135/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/523269a/ : SUCCESS in 55m 26s","accounts_in_message":[],"_revision_number":5},{"id":"290475171fb50e89591290a0bb860f7534dd1f24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 04:20:14.000000000","message":"Patch Set 5: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/544560a5f76e46a492bc222f3c040465 : FAILURE in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0cad0bf13c84e6ca00193208818f3f7 : SUCCESS in 12m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c60b749fff94c86b3b344ded2b5d901 : SUCCESS in 13m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd697a2ed4e9437c85e0781ceb235388 : RETRY_LIMIT in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aa52bfe833b40888808770d84107e3a : SUCCESS in 13m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/d6c69002ed144322ace5874be05c16c1 : SUCCESS in 54m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/50db15ff29ff49cc9d720d84478a4c77 : SUCCESS in 1h 02m 02s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cc422b29088f402490dae0e1a7320405 : SUCCESS in 1h 15m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17bd6ee13fba4c03b549abb437a28b1d : SUCCESS in 52m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a3d32703016b40918ebc8e250694233e : SUCCESS in 17m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7eaf1b00175c4091a24b9efec4ba2663 : SUCCESS in 6m 49s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4839a70cac5e47b9b16d28b2551e79c0 : SUCCESS in 9m 04s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/789b66e522864cebbf5fa102067a6f97 : SUCCESS in 7m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b06d26a99b5e4627ae201f1d4f2e3772 : SUCCESS in 8m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bc18edb80ed748db9eaf5762f26296a6 : SUCCESS in 49m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a8310c24c89541b191294aa582f7cad9 : SUCCESS in 6m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9635183fc0df41a29f11bba884c86d51 : SUCCESS in 1h 28m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f5a7d06880a645d6a31aa1ea025f3559 : SUCCESS in 1h 29m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3a3282bfd712455d8db3d78cd56e5bc1 : SUCCESS in 2h 03m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/abf4fa9f67db440881a79951d3d2deb4 : SUCCESS in 1h 10m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/387d9f983b6a4bf98645419dbd2a2472 : SUCCESS in 1h 30m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4284115cabde46938993f3d7ab80a3e3 : SUCCESS in 1h 01m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/12cf4a4358c24855ba4a7aa583d8af7e : SUCCESS in 1h 27m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dbc643b9ec9e406baf1b48923975c430 : SUCCESS in 2h 39m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e336e3b5aafc4e68bfe75bc2a98af750 : SUCCESS in 1h 06m 34s","accounts_in_message":[],"_revision_number":5},{"id":"aa1a043594c9eb93ab217d85e3462659ac07b3d9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-03-18 04:41:47.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/57a99baf88d14930b719102ed210ac2f : POST_FAILURE in 1h 48m 40s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/80a3fb8333604b49a339a6a1f314483b : POST_FAILURE in 3h 06m 15s","accounts_in_message":[],"_revision_number":5},{"id":"e6245affd2ad16726e3e1c2eb70e8c1629558636","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-18 04:47:22.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/35/817135/5/check/iscsi-driver-nimble/285c627/285c627.html : FAILURE in 1h 39m 00s","accounts_in_message":[],"_revision_number":5},{"id":"00b5cbb62222f9379315ed73b0f8b74cbc5a15c4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-18 05:55:39.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://93.100.254.243/nexenta/refs-changes-35-817135-5/ns5_nfs : SUCCESS in 4h 15m 0s\n* NexentaStor5_Cinder_iSCSI http://93.100.254.243/nexenta/refs-changes-35-817135-5/ns5_iscsi : SUCCESS in 4h 23m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"29bdd9d76018463ca520a0548dfc2afb37a623e6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-03-18 06:30:48.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job finger://zuulserver:17979/74748b1fb3114d428ed176a624188389 : RETRY_LIMIT in 1h 07m 17s","accounts_in_message":[],"_revision_number":5},{"id":"dd7f4f1042ecb2d96d7aff3e4a7075a32bb554f4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-18 07:49:32.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/35/817135/5/check/iscsi-driver-nimble/fbdb203/fbdb203.html : TIMED_OUT in 3h 00m 45s","accounts_in_message":[],"_revision_number":5},{"id":"35e72dd27b9d8c0746498e9c85b5639fd5e405fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2022-03-18 09:19:46.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"c4251269b42c5fc1b9bc1183d51ec468aee2bae9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 10:14:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c24f56f8c564c97b35a490f43d83326 : SUCCESS in 20m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/136e6de011b04106b327ec56416667a4 : SUCCESS in 28m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d1b02651e01f9af1dd1468ca0bf485c1e9a23572","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-18 10:17:51.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/817135 : FAILURE in 56m 37s","accounts_in_message":[],"_revision_number":6},{"id":"1551c07c4b315bc7ada4388c93d11b922cc329c9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-18 10:20:29.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/35/817135/6/check/iscsi-driver-nimble/ed948c5/ed948c5.html : FAILURE in 59m 24s","accounts_in_message":[],"_revision_number":6},{"id":"ac5aada2b4009f287ee5ddcb75f8c25600d04024","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-03-18 10:38:39.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/35/817135/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/43a8485/ : SUCCESS in 1h 10m 32s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/35/817135/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/bc0e599/ : SUCCESS in 1h 16m 37s","accounts_in_message":[],"_revision_number":6},{"id":"4e889421a3ea07f44ccfcdfcc8502550212e136a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-18 10:42:31.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/35/817135/6/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/72edc9e/72edc9e.html : SUCCESS in 1h 20m 45s","accounts_in_message":[],"_revision_number":6},{"id":"1b01f16ac40598799b4219078b2abc6b18a8582f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-03-18 10:43:10.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817135/6/iscsi/ : SUCCESS in 1h 22m 47s\n- cinder-smb http://cloudbase-ci.com/cinder/817135/6/smb/ : SUCCESS in 1h 21m 16s","accounts_in_message":[],"_revision_number":6},{"id":"b9eb5c85551439de8c88cfa74ccb5463245c5168","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-18 10:57:59.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/35/817135/6/check/iscsi-driver-alletra-6k/cf0ff36/cf0ff36.html : FAILURE in 1h 36m 59s","accounts_in_message":[],"_revision_number":6},{"id":"3c4b237ad1a04f8610e5b99cda9661252bbab90e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-18 11:22:54.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/35/817135/6/check/iscsi-driver-nimble/f1d5fde/f1d5fde.html : FAILURE in 1h 00m 57s","accounts_in_message":[],"_revision_number":6},{"id":"a552a2d4a802e33956e9319f94714cf0e715ddf7","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-03-18 11:47:21.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/35/817135/6/ : SUCCESS in 2h 19m 17s","accounts_in_message":[],"_revision_number":6},{"id":"84956e194cf08dbf2b88f7dcdd95e4b9e705ef4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 12:00:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88fe93b2549d41b4b24dbe096ea95490 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0db2da11ebab4c47b9d53f20e1152e17 : SUCCESS in 13m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9a0fc0bac9840389f121c90f068ac45 : SUCCESS in 12m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b6b7b7b2eb94380a51fff7dee5bd63f : RETRY_LIMIT in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1faa251b2bdf4eb7a762eb4588ac4c46 : SUCCESS in 13m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/35a864eb533446b3a32efd98f4f5dd51 : SUCCESS in 53m 20s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1304e4a9be4d4b1bb8d6706b636cf5b9 : SUCCESS in 1h 05m 37s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f3116e2c275447fa0262f5c7c8996ee : SUCCESS in 1h 02m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d945f0e7dd384a91b88818d68fb257c3 : SUCCESS in 53m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2cea58e873254ca39e932ee58220db3e : SUCCESS in 14m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2cad5ebd905b49f5b0c5eedcf66e3390 : SUCCESS in 7m 44s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/19152b9dd2db48358a7c70d2d5d31a66 : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bd10411399624b96b46b90145ef2afcb : SUCCESS in 8m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/496cceb30d304e72a1752b1e5f330dcc : SUCCESS in 9m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b64888e3dea34b1da2c86a3b118806ae : SUCCESS in 49m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b28106b3b0ad4a78890f8e904e36701b : SUCCESS in 6m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3736702ca9e94946ab754d3369b3156e : SUCCESS in 1h 18m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e4b569c19f0342989dfede9d57a789e0 : SUCCESS in 1h 46m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e1423300d8e146de9133e42b18b389e7 : SUCCESS in 1h 42m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ba4573496c7d42bf948d7231b44958c1 : SUCCESS in 1h 05m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8ed911e2682a41c0aa31dc547ee633b9 : SUCCESS in 1h 18m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6786d06da8554349b05085a3a9a1d3fb : SUCCESS in 45m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/99bd6a49fb9e4564a93fcea15a74b969 : SUCCESS in 1h 15m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fa8863be4df6425fac534981330e054f : SUCCESS in 2h 34m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f742d4f28a74b32bf8e70b089c9467d : SUCCESS in 48m 42s","accounts_in_message":[],"_revision_number":6},{"id":"251b18ec8cd65f2438b48b3d39508853916a7d20","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-03-18 12:21:27.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-35-817135-6/1647604842 : SUCCESS in 20m 43s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":6},{"id":"580b65e57a95fd0d303271899cfcad8877272e16","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-18 12:26:54.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/817135/817135/6/check-not-vote/ext-cinder-zuul/f7b0110 : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":6},{"id":"cf8c3323f16f1959a0b399460092a3b47f5568ac","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-03-18 12:32:00.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/6b4e82cf7d0e4eeb81f9be66282bb019 : POST_FAILURE in 1h 43m 16s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/8db79fa8475f40fe8390d8ca3080fb2b : TIMED_OUT in 3h 09m 28s","accounts_in_message":[],"_revision_number":6},{"id":"6983fe02ecc5d678b501f9287398c4b10af05e46","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-03-18 13:14:00.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job finger://zuulserver:17979/688c379435a245f1b0adf17469f0ad5f : RETRY_LIMIT in 1h 09m 40s","accounts_in_message":[],"_revision_number":6},{"id":"3ad59ea21c23ce426f31c6218dcfe0fccd650851","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-18 13:49:53.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://93.100.254.243/nexenta/refs-changes-35-817135-6/ns5_nfs : SUCCESS in 4h 29m 29s\n* NexentaStor5_Cinder_iSCSI http://93.100.254.243/nexenta/refs-changes-35-817135-6/ns5_iscsi : SUCCESS in 4h 17m 24s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"f3827d51fb0ec8975fe2e8504e43eae2479a77ea","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-19 04:49:43.000000000","message":"Patch Set 6:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15585/","accounts_in_message":[],"_revision_number":6},{"id":"b23b7f052840622e348be7163f2f645dfa4e36e1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-19 04:51:27.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15585/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"1330d56326e4d74dd23f85ca9c940c447ed3d615","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-03-19 13:24:40.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/35/817135/6/20220319190325/NEC-FC : SUCCESS in 3h 16m 06s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/35/817135/6/20220319190325/NEC-ISCSI : SUCCESS in 3h 21m 12s","accounts_in_message":[],"_revision_number":6},{"id":"88165b95c8da8245340aa7ca28d8b01cf540357e","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-03-22 05:00:20.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"344318ea633d7e14a049083dafa5faebffffdec1","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-03-29 15:13:25.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"35d83d43759595a6d959e9197ccec8ac970df2b5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-29 18:19:27.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0316a88f3657030a24dc39c3118f799761767382","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2022-03-30 00:43:35.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"9bf8b0a933805ff3564eba8140dd839e7c2108d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2022-03-30 00:43:38.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"421b27e125b3f35bbb60b203877b66bf8b13c7dd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-30 00:43:49.000000000","message":"Patch Set 7:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15662/","accounts_in_message":[],"_revision_number":7},{"id":"4b2314bf9efe896335e3894f4abe2a35a5f28de0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-30 00:45:31.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15662/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"7cbd185c00a4d131f8b6a6987f00a34e52af042d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 01:05:53.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/21f0368516fa47cda6b7afd5b4970260 : SUCCESS in 17m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40ae79e4478d42dcaf7a6cfd113c9d6a : SUCCESS in 20m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1b8beff5d18a0bcc4ff0c5e30e1ecd8660359d64","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-03-30 01:50:11.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817135/7/iscsi/ : SUCCESS in 56m 25s\n- cinder-smb http://cloudbase-ci.com/cinder/817135/7/smb/ : SUCCESS in 1h 05m 55s","accounts_in_message":[],"_revision_number":7},{"id":"b53cc30e070d7ee11901b0b5b84bb87d056d4157","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-03-30 02:02:55.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/35/817135/7/check/dsvm-tempest-cisco-zonemanager-job/475fa48/ : SUCCESS in 1h 11m 20s","accounts_in_message":[],"_revision_number":7},{"id":"b6d95adde30fead33144218a65c26e7845fcc3a6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-30 02:03:59.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/817135 : SUCCESS in 1h 18m 34s","accounts_in_message":[],"_revision_number":7},{"id":"fb1b00098415ba342b5ec34d3a7b9d2b7a6dba19","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-30 02:21:14.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/35/817135/7/check/iscsi-driver-alletra-6k/5c24ebf/5c24ebf.html : FAILURE in 1h 36m 32s","accounts_in_message":[],"_revision_number":7},{"id":"ce1dec433316ec7050934ffc256e621ed9300ca2","author":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"date":"2022-03-30 02:44:51.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e283ea40111276d799ef2bd6893fba89d986e6e5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-03-30 02:54:11.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/35/817135/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/e8bbc0b/ : FAILURE in 2h 08m 46s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/35/817135/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/e830345/ : SUCCESS in 1h 21m 15s","accounts_in_message":[],"_revision_number":7},{"id":"811411ee2642953ff437e9b73ff165f5dce0e084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 02:57:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56d11593c1ad47c49e58ea2a40b3faa4 : SUCCESS in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54b1e88a66744c038b782016139ee703 : SUCCESS in 12m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a85fcefe37840f2b8ea45a3b6e2f401 : SUCCESS in 13m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33c5a4b9ff0446b3a677ae74dd607607 : RETRY_LIMIT in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e4616968cb847deb6c0cbaed2cffdea : SUCCESS in 12m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/002bae763ad74c4bbb102e5297c3a44c : SUCCESS in 43m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fe880fcc18ef4518b67055feca4918b8 : SUCCESS in 1h 14m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56270345c409479b99a1f15930ad2318 : SUCCESS in 52m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c9d31140a8d3482ba32307081eb0c5e2 : SUCCESS in 15m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/07071c022d294d98a35465330a560220 : SUCCESS in 6m 57s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/708d066d63eb42a09b29da2d6bf72abd : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/622fd6f43df44090b5104d2b8c716a43 : SUCCESS in 9m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9c70f42f7a30476abf190440d8fa4621 : SUCCESS in 8m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/614cb0761b4744c28887d3d11cc8899b : SUCCESS in 42m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/63ddceead1e94bbb84776317dcc770c3 : SUCCESS in 6m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/70230e0d4ced403e822655fd2ffa0225 : SUCCESS in 1h 08m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ea98a0d228e444ab99399b9fe6b8dde9 : SUCCESS in 1h 35m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/90a5b9d843e147ac913ab4538f35d35a : SUCCESS in 1h 20m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3528555d7c6f44faae162e4cc9e3cb40 : SUCCESS in 1h 08m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/88ae31bbf0dd48a3b133adafec978b07 : SUCCESS in 1h 37m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bfc51d7467be42e4a74fa32d0654cc8c : SUCCESS in 42m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c766e44994d943cea15c7ffabac4fcce : SUCCESS in 1h 32m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0fe15b1b387e4ce090ac5706bf181858 : SUCCESS in 2h 11m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e95d3f3d7284045929d07183aeeef5f : SUCCESS in 39m 48s","accounts_in_message":[],"_revision_number":7},{"id":"cb6115411b7dfaf4f7008e6883cf6dd6f81fb1fc","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-03-30 02:58:30.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/817135/7/upstream-solidfire-check/cinder-solidfire-iSCSI/a89bb99/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"7dc93f490ab536f73f021cd7ef2491d7e35dc906","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-30 03:01:59.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/817135/7/upstream-check/cinder-cDOT-iSCSI/231e0c4/ : FAILURE in 3m 47s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/817135/7/upstream-check/cinder-cDOT-NFS/4a6b347/ : FAILURE in 2m 27s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/817135/7/upstream-check/cinder-cDOT-FCP/3e726c2/ : FAILURE in 3m 05s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"1d79ad22e2d4d0f47fc39ee2e3dfa0c0fcb2a00b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-03-30 03:14:26.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/59d189b793214022a00c94dae665aaff : POST_FAILURE in 1h 45m 57s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/2768ea89d41f4d099753c443f6d3a4a0 : POST_FAILURE in 2h 21m 09s","accounts_in_message":[],"_revision_number":7},{"id":"640a59c42b6cdac9d8cd97373660a91018f922eb","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-30 03:15:16.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"7ae90e0282fd9b8c054134a93da8da4a3ac08266","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-30 03:16:02.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a7c559dfffa51b5efc38306beeda9b0649d8b662","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-03-30 03:48:43.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/35/817135/7/ : SUCCESS in 2h 04m 55s","accounts_in_message":[],"_revision_number":7},{"id":"ab01f47eb672fffc4f230db0af3581d80da7d26a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-30 04:02:06.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-35-817135-7/ns5_nfs : SUCCESS in 3h 17m 59s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-35-817135-7/ns5_iscsi : SUCCESS in 3h 17m 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"445cf0787ef4454edc363ead953583d3da3d87ad","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-03-30 04:18:22.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/35/817135/7/check/DellEMC_PowerFlex/b933be8/DellEMC_PowerFlex/507 : SUCCESS in 1h 20m 55s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"7af1d2f021dc5b59ae00582cec24de281498a32d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-03-30 04:21:25.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/35/817135/7/check/DellEMC_PowerStore_iSCSI/4e1a139/DellEMC_PowerStore_iSCSI/392 : SUCCESS in 1h 23m 57s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/35/817135/7/check/DellEMC_PowerStore_FC/044bbc6/DellEMC_PowerStore_FC/408 : SUCCESS in 1h 10m 28s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"0188f08e76dae3fedeee1fcf5349681537ea25ad","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-03-30 05:02:50.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/35/817135/7/check/DellEMC_SC_FC/084fb14/DellEMC_SC_FC/377 : SUCCESS in 1h 04m 31s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/35/817135/7/check/DellEMC_SC_iSCSI/0c3c730/DellEMC_SC_iSCSI/399 : SUCCESS in 2h 05m 26s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"a7ad3e03dd79d4693d38dad853decf8235d430c4","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-03-30 05:05:10.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/817135-7/HPMSA-iSCSI/ : SUCCESS in 50:26.10\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/817135-7/HPMSA-FC/ : SUCCESS in 44:36.88","accounts_in_message":[],"_revision_number":7},{"id":"80854764cbde685887cb32b450c45fc8c6f14988","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-30 05:20:29.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera iscsi-driver-primera : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"8615a2e033a6bf01532a6f6df90d70878c52a55e","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-03-30 08:04:48.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/817135-7/Lenovo-iSCSI/ : SUCCESS in 1:01:41\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/817135-7/Lenovo-FC/ : SUCCESS in 59:01.22","accounts_in_message":[],"_revision_number":7},{"id":"7c748910b2e82c74605f7ddf971159fe18cf354a","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-03-30 09:01:58.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/817135-7/PVME-iSCSI/ : SUCCESS in 1:05:04\n- PVME-FC https://os-logs.seagate.com/dell-ci/817135-7/PVME-FC/ : SUCCESS in 1:04:23","accounts_in_message":[],"_revision_number":7},{"id":"c649280a2a64b932d7a3b82e8ab5174dcd82d234","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-03-30 09:31:04.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/817135-7/Seagate-iSCSI/ : SUCCESS in 1:09:47\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/817135-7/Seagate-FC/ : SUCCESS in 1:02:51","accounts_in_message":[],"_revision_number":7},{"id":"4dc9b40e82675d5040ede8c431e0004be672e3f0","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-03-30 12:26:52.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/35/817135/7/check/DellEMC_PowerMAX_FC/28915db/DellEMC_PowerMAX_FC/395 : SUCCESS in 3h 01m 57s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/35/817135/7/check/DellEMC_PowerMAX_iSCSI/c2a1418/DellEMC_PowerMAX_iSCSI/397 : FAILURE in 4m 30s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":7},{"id":"469361f26c81945a84f1235333f43d83bad80b94","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-03-30 15:43:47.000000000","message":"Patch Set 7:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/35/817135/7/20220330235036/NEC-V-ISCSI : FAILURE in 49m 01s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/35/817135/7/20220330235036/NEC-V-FC : FAILURE in 53m 08s","accounts_in_message":[],"_revision_number":7},{"id":"2b63c8defa5c839cab2e422cf7d08d3bc1e4ad0a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-30 18:38:38.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/35/817135/7/check/iscsi-driver-nimble/f0b9079/f0b9079.html : FAILURE in 1h 35m 35s","accounts_in_message":[],"_revision_number":7},{"id":"35f15d1b13319559666d37af1828fb5fa7b3a9bc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-30 20:15:44.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/35/817135/7/check/iscsi-driver-nimble/5ef8a55/5ef8a55.html : FAILURE in 1h 35m 33s","accounts_in_message":[],"_revision_number":7},{"id":"f3b731033db4e48afbb5cfec082d3ae9dc7ee705","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-03-31 12:03:16.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a41b9c0a69a310c96bdea7acc48b748c17bc431b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-03-31 15:38:56.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/35/817135/7/20220331211728/NEC-FC : SUCCESS in 3h 19m 27s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/35/817135/7/20220331211728/NEC-ISCSI : SUCCESS in 3h 21m 25s","accounts_in_message":[],"_revision_number":7},{"id":"c5549680b53d519b0ff0e88e7823f0bdbf570f5f","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2022-04-04 09:28:16.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"fa782c95753c790da024dfa7f5fe8d452f8e5258","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-04-06 13:35:34.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"81804354ad34a08111c6883b4053d5ac0acb4999","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-04-06 13:36:47.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9d5f5f93e69b016fff6653b3a6eeaec4abec9268","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 14:17:08.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/abaeddbc9bff4e1d9efaed6c538200d6 : SUCCESS in 25m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f494e87996440d583699382dfe96f40 : SUCCESS in 23m 37s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fa60c0f2f58361a2047c05ea00d9ca6ddf75352f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-06 15:03:11.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/817135/817135/8/check-not-vote/ext-cinder-zuul/e2d0843 : SUCCESS in 43m 35s","accounts_in_message":[],"_revision_number":8},{"id":"79f1f962576bc34258400d24da1a4d080c925158","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-06 15:51:50.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817135/8/iscsi/ : FAILURE in 1h 45m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/817135/8/smb/ : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":8},{"id":"815082af82522ad59908052305af7a37d04adf41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 16:07:00.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71452dbc6e704ea6a23e2481afbf31b0 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87da8cdeeb1f48a39671d6111f79c241 : SUCCESS in 10m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fb18f1ba0124c3488f9139196207020 : SUCCESS in 12m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c26eed18a1e4dc1935d212ac4a35193 : RETRY_LIMIT in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a90e86fe9596462a9022a8e7788beafc : SUCCESS in 11m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/8b693036ac4e42da94b97d4ad544b962 : SUCCESS in 59m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b6ff214a4174aaf96b73d68c3d70506 : SUCCESS in 1h 20m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d80f218772514b3587994c3f2c56ec4c : SUCCESS in 52m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1cb85c477d9448588e60c17627ee2715 : SUCCESS in 15m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9f429d617043437bae14acb1ec2adb79 : SUCCESS in 5m 53s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/42366094f1c84319ab36ff9c573c65db : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dec7240a10d1439d80b9617e62ec45f5 : SUCCESS in 6m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0f6d5fc51d9d49e88e8ccc6ef76eacbd : SUCCESS in 8m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/563731124287467894457473af94ec46 : SUCCESS in 33m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/62741ae4026e4cf1b593fe23feeb9e6b : FAILURE in 4m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/16df70b086e34e44a1ef40b432191e15 : SUCCESS in 1h 02m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e1616ec824ef49c88750e8c36bdd77bc : SUCCESS in 1h 10m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3b58e43e7a9b46cab5fc531672e8df7c : SUCCESS in 1h 50m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/298dfa38fbbf46a1bc3c67ebc2ab0e3b : SUCCESS in 1h 02m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a002ac5189964a06bc1330a7b7aba1b1 : SUCCESS in 1h 27m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8e666253aa774316ab7a1dab7fefabfe : SUCCESS in 30m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6e2dbe559d60420ca6c1f2d8fe573ec2 : SUCCESS in 1h 23m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ca6893d9a6b5477285aaee5f20690147 : SUCCESS in 2h 24m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e54f360a47749219b11143f351e5268 : SUCCESS in 1h 18m 16s","accounts_in_message":[],"_revision_number":8},{"id":"1a1a011392a012b28ccbc0c09331269fab7ab66a","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-04-06 16:08:50.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/817135/8/upstream-solidfire-check/cinder-solidfire-iSCSI/9a7f0fa/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"7287d2cdc1659ef965a076f116a3173d280c961f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-04-06 16:22:42.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/817135/8/upstream-check/cinder-cDOT-iSCSI/70170d1/ : FAILURE in 4m 20s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/817135/8/upstream-check/cinder-cDOT-NFS/c41b332/ : FAILURE in 14m 10s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/817135/8/upstream-check/cinder-cDOT-FCP/f1310bf/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"1286d98662f83f2ed81a6a6d3d98680e6983134c","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-04-06 16:41:28.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-35-817135-8/1649261220 : SUCCESS in 17m 40s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":8},{"id":"6ed35f99733f5aad7f9eab61c5373aef67ff9bae","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-06 17:01:12.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-35-817135-8/ns5_nfs : SUCCESS in 3h 25m 7s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-35-817135-8/ns5_iscsi : SUCCESS in 3h 22m 44s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"4c1c53c1d4532cb2713f21be251a940eadd68f88","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-06 17:03:59.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/817135 : FAILURE in 54m 57s","accounts_in_message":[],"_revision_number":8},{"id":"4751959e77e6180cda68ada7ce2de394b5992d67","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-06 17:53:12.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/35/817135/8/check/iscsi-driver-alletra-6k/a71394a/a71394a.html : FAILURE in 1h 02m 00s","accounts_in_message":[],"_revision_number":8},{"id":"8c3feffce2cbd895c8e09d85a26d42e74b85c051","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-06 18:30:55.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/817135 : SUCCESS in 1h 19m 44s","accounts_in_message":[],"_revision_number":8},{"id":"d9a53697e3849c3e13584de2cf334ff508c84b97","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-06 18:49:26.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-35-817135-8/ns5_nfs : ABORTED in 1h 31m 52s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-35-817135-8/ns5_iscsi : ABORTED in 1h 32m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"b83b2dd3f2244bbf5edacd75bcf85b2573589bcd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-04-06 19:52:46.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/35/817135/8/check/DellEMC_PowerFlex/7c3e022/DellEMC_PowerFlex/537 : SUCCESS in 1h 28m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"a910f592325a17035e0d10e966e0cce2783c85d1","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-04-06 21:10:19.000000000","message":"Patch Set 8:\n\nBuild failed.","accounts_in_message":[],"_revision_number":8},{"id":"b4132a7a51fced7c6e242119eb46b92c733185c2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-06 21:12:51.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/35/817135/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/87e5fe6/ : FAILURE in 2h 45m 49s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/35/817135/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/4b49576/ : SUCCESS in 1h 22m 52s","accounts_in_message":[],"_revision_number":8},{"id":"9bdf1b3c5145e78c4a4e78c2e907fa014a152e5e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-06 21:36:10.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/049ded0161234846a80cddebc8d44841 : POST_FAILURE in 2h 16m 36s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"84b22c230de9f7e74118cdfc2600b341c19014ee","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-04-06 22:29:49.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/35/817135/8/check/DellEMC_SC_FC/d14876c/DellEMC_SC_FC/405 : SUCCESS in 1h 05m 00s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/35/817135/8/check/DellEMC_SC_iSCSI/91fc265/DellEMC_SC_iSCSI/426 : SUCCESS in 2h 01m 29s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":8},{"id":"4712db904c3d38796695c1f2716a0b26bad52fec","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-04-07 01:18:24.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/817135-8/Lenovo-iSCSI/ : SUCCESS in 1:05:18\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/817135-8/Lenovo-FC/ : SUCCESS in 57:55.28","accounts_in_message":[],"_revision_number":8},{"id":"32cda37bdb467f959df5449e49c9f83d84f0ce69","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-04-07 01:51:55.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/817135-8/PVME-iSCSI/ : SUCCESS in 1:05:11\n- PVME-FC https://os-logs.seagate.com/dell-ci/817135-8/PVME-FC/ : SUCCESS in 1:04:07","accounts_in_message":[],"_revision_number":8},{"id":"43a03a000f108e0838d8e2694099d3a0899b9d04","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-04-07 02:00:53.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c86ba31882994d3cb84c7dfd72f8cdd4 : SUCCESS in 1h 19m 13s","accounts_in_message":[],"_revision_number":8},{"id":"23b5d8d41d93343a5b38e0e31e126b95f0e3ece1","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-04-07 02:17:18.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/817135-8/Seagate-iSCSI/ : SUCCESS in 1:07:05\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/817135-8/Seagate-FC/ : SUCCESS in 1:04:01","accounts_in_message":[],"_revision_number":8},{"id":"02b83abc10adec57df5c4b021f5bfc511cc0cd87","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-07 02:30:36.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/35/817135/8/check/dsvm-tempest-cisco-zonemanager-job/b34fd51/ : SUCCESS in 1h 13m 47s","accounts_in_message":[],"_revision_number":8},{"id":"dde76a5fb22be41ba8f35ecff009a44cb7db142f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-07 03:50:55.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-35-817135-8/ns5_nfs : SUCCESS in 3h 24m 7s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-35-817135-8/ns5_iscsi : SUCCESS in 3h 25m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"84ca5be650de0166cfcdf5f8c39cf55e613660a3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-07 14:05:51.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/35/817135/8/check/iscsi-driver-nimble/f303e2d/f303e2d.html : FAILURE in 1h 00m 35s","accounts_in_message":[],"_revision_number":8},{"id":"627f1396fb307dfea80f4bcc7ed3d23355741369","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-07 15:09:06.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/35/817135/8/check/iscsi-driver-nimble/5d0dffd/5d0dffd.html : FAILURE in 1h 01m 50s","accounts_in_message":[],"_revision_number":8},{"id":"066ebdd6999aed6b3bf3c2c87b66a37be0718d18","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-04-08 03:03:04.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/35/817135/8/ : FAILURE in 2h 04m 06s","accounts_in_message":[],"_revision_number":8},{"id":"f1beeaca849e0292aab6c4760b8c819a8e52f855","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-04-08 03:20:15.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/35/817135/8/20220408112430/NEC-V-ISCSI : FAILURE in 52m 04s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/35/817135/8/20220408112430/NEC-V-FC : FAILURE in 55m 42s","accounts_in_message":[],"_revision_number":8},{"id":"f7cd1397869ceda45575218b6d592e6c8156f33e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-04-08 11:40:15.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/35/817135/8/20220408171726/NEC-FC : SUCCESS in 3h 21m 31s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/35/817135/8/20220408171726/NEC-ISCSI : SUCCESS in 3h 22m 45s","accounts_in_message":[],"_revision_number":8},{"id":"a686165847c2a9dc96de8afc77c9d507a84e95ed","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-04-11 10:08:03.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/35/817135/8/check/DellEMC_XtremIO_FC/fb538bf/DellEMC_XtremIO_FC/929 : FAILURE in 15m 01s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/35/817135/8/check/DellEMC_XtremIO_iSCSI/1ba420a/DellEMC_XtremIO_iSCSI/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"ecb9f0fe9e0d643d47343740b3b8e4508558bee8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-06-17 16:23:40.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"26d66dda6f6345f75c95b8b64463be3ddbefb933","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-06-17 16:24:04.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"962bf9163c860708350f1e64fb6ceeeb545f6712","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-07-11 10:35:40.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ee12f12c7c1e5baa7586f1ba30a7ee8444140a35","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2023-01-06 00:29:39.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"839f1612bee032fd69de4db4f96c890d51a35d4a","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2023-04-05 05:31:27.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fa500b6b234f4d4db4bb99e6418bdb24c589cf80","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2025-10-31 15:44:55.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"add3668d39ec8a7f6c36eb3642347676f86a2f1e","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-11-07 15:45:52.000000000","message":"Patch Set 8: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"724a1095e27b4564a9fdbdb8f9cc9611158c3f2e","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2025-11-10 08:06:46.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"04c11005461700dec84531751b0b99f2b65cc387","revisions":{"d8931d0719f3e187b3f67d2923e6f1d4fee3c430":{"kind":"REWORK","_number":1,"created":"2021-11-09 03:44:51.000000000","uploader":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"ref":"refs/changes/35/817135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/817135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/817135/1"}}},"commit":{"parents":[{"commit":"bada2f2487116fcdeedc38b6326400ca9684a6e6","subject":"Merge \"Fixed copy-on-write mode in GPFS NFS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bada2f2487116fcdeedc38b6326400ca9684a6e6"}]}],"author":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-11-09 03:35:46.000000000","tz":540},"committer":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-11-09 03:35:46.000000000","tz":540},"subject":"Retry _get_ig() when IG already exists.","message":"Retry _get_ig() when IG already exists.\n\nThis fix will handle an exception from XMS, ig_name_not_unique, when\nconcurrent initialize_connection() runs.\n\nPreviously, concurrent mulitple live-migration to the node which is not\nregistered to XtremIO yet caused the error due to the multiple\n_create_ig() were executed.\n\nChange-Id: I0c6cd0bc7741656bee004c3e3c74f772e5691202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d8931d0719f3e187b3f67d2923e6f1d4fee3c430"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d8931d0719f3e187b3f67d2923e6f1d4fee3c430"}]},"branch":"refs/heads/master"},"e2f27e95aeddcc71b5ba28e4919ccf0932093a73":{"kind":"REWORK","_number":2,"created":"2021-11-09 04:16:58.000000000","uploader":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"ref":"refs/changes/35/817135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/817135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/817135/2"}}},"commit":{"parents":[{"commit":"bada2f2487116fcdeedc38b6326400ca9684a6e6","subject":"Merge \"Fixed copy-on-write mode in GPFS NFS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bada2f2487116fcdeedc38b6326400ca9684a6e6"}]}],"author":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-11-09 03:35:46.000000000","tz":540},"committer":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-11-09 04:16:47.000000000","tz":540},"subject":"Retry _get_ig() when IG already exists.","message":"Retry _get_ig() when IG already exists.\n\nThis fix will handle an exception from XMS, ig_name_not_unique, when\nconcurrent initialize_connection() runs.\n\nPreviously, concurrent mulitple live-migration to the node which is not\nregistered to XtremIO yet caused the error due to the multiple\n_create_ig() were executed.\n\nChange-Id: I0c6cd0bc7741656bee004c3e3c74f772e5691202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e2f27e95aeddcc71b5ba28e4919ccf0932093a73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e2f27e95aeddcc71b5ba28e4919ccf0932093a73"}]},"branch":"refs/heads/master"},"92bd93ded5b31c70f0088a4b7c673f7512fc3c44":{"kind":"REWORK","_number":3,"created":"2021-11-09 08:07:01.000000000","uploader":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"ref":"refs/changes/35/817135/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/817135/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/817135/3"}}},"commit":{"parents":[{"commit":"bada2f2487116fcdeedc38b6326400ca9684a6e6","subject":"Merge \"Fixed copy-on-write mode in GPFS NFS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bada2f2487116fcdeedc38b6326400ca9684a6e6"}]}],"author":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-11-09 03:35:46.000000000","tz":540},"committer":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-11-09 08:06:49.000000000","tz":540},"subject":"Retry _get_ig() when IG already exists.","message":"Retry _get_ig() when IG already exists.\n\nThis fix will handle an exception from XMS, ig_name_not_unique, when\nconcurrent initialize_connection() runs.\n\nPreviously, concurrent mulitple live-migration to the node which is not\nregistered to XtremIO yet caused the error due to the multiple\n_create_ig() were executed.\n\nChange-Id: I0c6cd0bc7741656bee004c3e3c74f772e5691202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92bd93ded5b31c70f0088a4b7c673f7512fc3c44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92bd93ded5b31c70f0088a4b7c673f7512fc3c44"}]},"branch":"refs/heads/master"},"e5b333b542d5b9d4b054fe4becfac73789bb3206":{"kind":"REWORK","_number":4,"created":"2021-11-10 04:24:32.000000000","uploader":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"ref":"refs/changes/35/817135/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/817135/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/817135/4"}}},"commit":{"parents":[{"commit":"bada2f2487116fcdeedc38b6326400ca9684a6e6","subject":"Merge \"Fixed copy-on-write mode in GPFS NFS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bada2f2487116fcdeedc38b6326400ca9684a6e6"}]}],"author":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-11-09 03:35:46.000000000","tz":540},"committer":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-11-10 04:24:20.000000000","tz":540},"subject":"Retry _get_ig() when IG already exists.","message":"Retry _get_ig() when IG already exists.\n\nThis fix will handle an exception from XMS, ig_name_not_unique, when\nconcurrent initialize_connection() runs.\n\nPreviously, concurrent mulitple live-migration to the node which is not\nregistered to XtremIO yet caused the error due to the multiple\n_create_ig() were executed.\n\nChange-Id: I0c6cd0bc7741656bee004c3e3c74f772e5691202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e5b333b542d5b9d4b054fe4becfac73789bb3206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e5b333b542d5b9d4b054fe4becfac73789bb3206"}]},"branch":"refs/heads/master"},"d3971a1b65da798c0f66b1f93c6fb0c9f6ad6327":{"kind":"REWORK","_number":5,"created":"2022-03-18 01:31:56.000000000","uploader":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"ref":"refs/changes/35/817135/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/817135/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/817135/5"}}},"commit":{"parents":[{"commit":"bada2f2487116fcdeedc38b6326400ca9684a6e6","subject":"Merge \"Fixed copy-on-write mode in GPFS NFS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bada2f2487116fcdeedc38b6326400ca9684a6e6"}]}],"author":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-11-09 03:35:46.000000000","tz":540},"committer":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2022-03-18 01:31:56.000000000","tz":0},"subject":"Retry _get_ig() when IG already exists.","message":"Retry _get_ig() when IG already exists.\n\nThis fix will handle an exception from XMS, ig_name_not_unique, when\nconcurrent initialize_connection() runs.\n\nPreviously, concurrent mulitple live-migration to the node which is not\nregistered to XtremIO yet caused the error due to the multiple\n_create_ig() were executed.\n\nChange-Id: I0c6cd0bc7741656bee004c3e3c74f772e5691202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3971a1b65da798c0f66b1f93c6fb0c9f6ad6327"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3971a1b65da798c0f66b1f93c6fb0c9f6ad6327"}]},"branch":"refs/heads/master"},"7471e53d4d87e7081e5324ee26f6fa513a9652ef":{"kind":"REWORK","_number":6,"created":"2022-03-18 09:19:46.000000000","uploader":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"ref":"refs/changes/35/817135/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/817135/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/817135/6"}}},"commit":{"parents":[{"commit":"bada2f2487116fcdeedc38b6326400ca9684a6e6","subject":"Merge \"Fixed copy-on-write mode in GPFS NFS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bada2f2487116fcdeedc38b6326400ca9684a6e6"}]}],"author":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-11-09 03:35:46.000000000","tz":540},"committer":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2022-03-18 09:19:46.000000000","tz":0},"subject":"Retry _get_ig() when IG already exists.","message":"Retry _get_ig() when IG already exists.\n\nThis fix will handle an exception from XMS, ig_name_not_unique, when\nconcurrent initialize_connection() runs.\n\nPreviously, concurrent mulitple live-migration to the node which is not\nregistered to XtremIO yet caused the error due to the multiple\n_create_ig() were executed.\n\nChange-Id: I0c6cd0bc7741656bee004c3e3c74f772e5691202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7471e53d4d87e7081e5324ee26f6fa513a9652ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7471e53d4d87e7081e5324ee26f6fa513a9652ef"}]},"branch":"refs/heads/master"},"62b3fbe134793c91f7da792b8b30929e4189a57e":{"kind":"REWORK","_number":7,"created":"2022-03-30 00:43:38.000000000","uploader":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"ref":"refs/changes/35/817135/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/817135/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/817135/7"}}},"commit":{"parents":[{"commit":"bada2f2487116fcdeedc38b6326400ca9684a6e6","subject":"Merge \"Fixed copy-on-write mode in GPFS NFS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bada2f2487116fcdeedc38b6326400ca9684a6e6"}]}],"author":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-11-09 03:35:46.000000000","tz":540},"committer":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2022-03-30 00:43:05.000000000","tz":0},"subject":"Retry _get_ig() when IG already exists.","message":"Retry _get_ig() when IG already exists.\n\nThis fix will handle an exception from XMS, ig_name_not_unique, when\nconcurrent initialize_connection() runs.\n\nPreviously, concurrent mulitple live-migration to the node which is not\nregistered to XtremIO yet caused the error due to the multiple\n_create_ig() were executed.\n\nChange-Id: I0c6cd0bc7741656bee004c3e3c74f772e5691202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/62b3fbe134793c91f7da792b8b30929e4189a57e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/62b3fbe134793c91f7da792b8b30929e4189a57e"}]},"branch":"refs/heads/master"},"04c11005461700dec84531751b0b99f2b65cc387":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2022-04-06 13:35:34.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/35/817135/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/35/817135/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/35/817135/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/35/817135/8"}}},"commit":{"parents":[{"commit":"bada2f2487116fcdeedc38b6326400ca9684a6e6","subject":"Merge \"Fixed copy-on-write mode in GPFS NFS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bada2f2487116fcdeedc38b6326400ca9684a6e6"}]}],"author":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2021-11-09 03:35:46.000000000","tz":540},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2022-04-06 13:35:34.000000000","tz":0},"subject":"Retry _get_ig() when Initiator Group already exists.","message":"Retry _get_ig() when Initiator Group already exists.\n\nThis fix will handle an exception from XMS, ig_name_not_unique, when\nconcurrent initialize_connection() runs.\n\nPreviously, concurrent mulitple live-migration to the node which is not\nregistered to XtremIO yet caused the error due to the multiple\n_create_ig() were executed.\n\nChange-Id: I0c6cd0bc7741656bee004c3e3c74f772e5691202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04c11005461700dec84531751b0b99f2b65cc387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04c11005461700dec84531751b0b99f2b65cc387"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"failing_atoms":[],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
