)]}'
{"id":"openstack%2Fcinder~817417","triplet_id":"openstack%2Fcinder~stable%2Fussuri~I8287460b902dd525aa5313861142f5fb8490e60a","project":"openstack/cinder","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2021-11-11 14:38:30.000000000","reason":"Change was submitted"},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2021-11-11 14:38:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8287460b902dd525aa5313861142f5fb8490e60a","subject":"RBD: Open RBD images read-only where possible","status":"MERGED","created":"2021-11-10 15:01:04.000000000","updated":"2021-11-18 08:47:19.000000000","submitted":"2021-11-11 14:38:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817417","cherry_pick_of_change":817261,"cherry_pick_of_patch_set":1,"meta_rev_id":"7006e73eff199dfbd420e06ae2c64bd1e4d655ac","_number":817417,"virtual_id_number":817417,"owner":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-11 14:38:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-11-11 00:23:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-11 05:05:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-11 05:05:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-11-11 00:23:05.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-10 15:01:04.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-11-10 15:26:25.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-11-10 15:54:06.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-10 16:02:14.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-11-10 16:37:19.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-10 17:23:16.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-11-10 18:46:12.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-10 20:35:43.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2021-11-10 22:56:28.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-10 22:58:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-10 23:12:26.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-11-10 23:50:14.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-11 00:23:05.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-11-11 02:38:02.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-11 05:05:54.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-11-11 11:44:02.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2021-11-18 07:38:25.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"}],"messages":[{"id":"6867bf05ac5448e4f660296255c2e235c7d30947","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-11-10 15:01:04.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"5aaef66166af4e6e759e4823ea7cf8484369a93e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-10 15:26:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/17/817417/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/14fa3c4/ : FAILURE in 8m 01s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/17/817417/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6adc892/ : FAILURE in 24m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1b899713cc7d59260c248ad9b1801109d00f5bcf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-10 15:54:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/17/817417/1/check/iscsi-driver-primera/6b181af/6b181af.html : FAILURE in 51m 04s","accounts_in_message":[],"_revision_number":1},{"id":"eb69c8be3579afcec2ac3095f3866dabf242baca","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-10 16:02:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817417/1/iscsi/ : SUCCESS in 59m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/817417/1/smb/ : SUCCESS in 1h 00m 33s","accounts_in_message":[],"_revision_number":1},{"id":"cde520eaaa8cab6af351c998fa34d9d1a3d347a6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-10 16:14:39.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/17/817417/1/check/iscsi-driver-alletra-6k/d39af43/d39af43.html : FAILURE in 51m 58s","accounts_in_message":[],"_revision_number":1},{"id":"06e83293bf1b880997a8af6bfbda73b2c7f7f238","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-10 16:33:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/17/817417/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/522f644/522f644.html : FAILURE in 49m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e6795804beb9323358fe81f30dab2e0989ff093f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-10 16:37:19.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/17/817417/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"e822aa43c1c0c9e4b7f3db09cdd2d364f08472ce","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-10 17:23:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/17/817417/1/check/dsvm-tempest-cisco-zonemanager-job/0790d44/ : FAILURE in 31m 38s","accounts_in_message":[],"_revision_number":1},{"id":"fc85715b178c3faa01673c4cddd6fc27ab62be99","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-10 17:33:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/17/817417/1/check/iscsi-driver-nimble/1db5d8c/1db5d8c.html : FAILURE in 59m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e740a991556cc5b21bdf9f9adbcaff9058889e12","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-10 18:46:12.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic tempest-dsvm-full-bionic : NODE_FAILURE in 0s\n- tempest-dsvm-full-bionic-py3 tempest-dsvm-full-bionic-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-bionic-py3 finger://zuul-v3-executor.novalocal/6b86d87673634cdb81e57b4549fd6081 : TIMED_OUT in 3h 10m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f6bcec4f06478e99caf5eb5676fd39e0e453fdea","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-11-10 20:35:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"611712e55bacc26e7720630cf8a65f2c7b68c4a4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-10 20:37:10.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/17/817417/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"73fa37de690b39a5f18f48ae26bc87dfce07e1df","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-10 21:41:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817417/1/iscsi/ : SUCCESS in 1h 02m 58s\n- cinder-smb http://cloudbase-ci.com/cinder/817417/1/smb/ : SUCCESS in 1h 04m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2c960e591461808be6f4ede2695602692ad1f1ec","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-10 22:56:28.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-817417-1/ns5_nfs : SUCCESS in 4h 51m 0s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-817417-1/ns5_iscsi : SUCCESS in 4h 23m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"38750548901250a950cc0c2246b749cad0501f07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 22:58:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1141d23b4b534807a50aba472cea24ff : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b46df0b61fe4ae38d3284fe3ae96b1a : SUCCESS in 10m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d44800380c645e989cf54dd8814160d : SUCCESS in 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f370fb2c4eae440885c0763e55f60cfc : SUCCESS in 14m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/5ea946b141ed4a37a6012edf70835ce0 : SUCCESS in 1h 11m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2266b852b0d040d1b4d34154f681ae13 : SUCCESS in 1h 23m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0d2095d13f945cfb84de43f67a7610d : SUCCESS in 8m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/924b9d1d8c674e4e8edf62ad32910ad9 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6ad1a14a2a3441b3b0e7c0ed51a58885 : SUCCESS in 8m 24s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/12ffd8c6b0484168916051e7e17fa448 : SUCCESS in 8m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8cefa4b089fc4fdc97858b3524cfa1f7 : SUCCESS in 58m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/10887f00edb7444d89300f20ab29d4c7 : SUCCESS in 9m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/018330bc60da4b2eb353471f5f40024d : SUCCESS in 1h 26m 02s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/094183b50b6a4646a76df7d8a906247f : SUCCESS in 1h 31m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c4674457560247cd9de92bd76d7db73f : SUCCESS in 1h 28m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f885176dc727430e95944cf5d4db0a38 : SUCCESS in 1h 25m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/98d7761386bd414aa0a19339ea5828f8 : FAILURE in 1h 05m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2afc4f1b4ee044c7a4b29f08047d3ce7 : SUCCESS in 2h 19m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9026900795914347a8b29753431ff4a1 : SUCCESS in 55m 47s","accounts_in_message":[],"_revision_number":1},{"id":"33d414e539ea8256714b4da6e853c52b2434ef58","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-11-10 23:12:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/17/817417/1/upstream-solidfire-check/cinder-solidfire-iSCSI/83b319f/ : FAILURE in 12m 39s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"1cd3e6d984c2f25b375da82106bc3efd33c9487e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-10 23:50:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/17/817417/1/upstream-check/cinder-cDOT-iSCSI/5b8599e/ : FAILURE in 50m 50s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/17/817417/1/upstream-check/cinder-cDOT-NFS/101efb3/ : FAILURE in 9m 10s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/17/817417/1/upstream-check/cinder-cDOT-FCP/1c38c95/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"747b6c181fa75560d04e3e86483371c936beae95","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-11 00:06:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/17/817417/1/check/dsvm-tempest-cisco-zonemanager-job/06fbcad/ : FAILURE in 31m 48s","accounts_in_message":[],"_revision_number":1},{"id":"58d035df21061391cb090f1d91e93db7095dc9ba","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-11-11 00:23:05.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"21b99bf3786df9250f35eab903040e322057bdba","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-11 00:25:07.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/6086434c7aa7479f9287c0b898a15733 : POST_FAILURE in 3h 03m 11s\n- tempest-dsvm-full-bionic-py3 tempest-dsvm-full-bionic-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-bionic-py3 finger://zuul-v3-executor.novalocal/7140eba7b05f42c0b5e330d4dcfca093 : POST_FAILURE in 29m 43s","accounts_in_message":[],"_revision_number":1},{"id":"dc8f84042830245135ccf4dfbd4738dacac014bb","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-11 02:38:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/17/817417/1/20211111085113/NEC-ISCSI : SUCCESS in 2h 43m 54s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/17/817417/1/20211111085113/NEC-FC : SUCCESS in 2h 46m 46s","accounts_in_message":[],"_revision_number":1},{"id":"fec91954420a448bbdef3bde8e2f93ab94ff837c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-11 04:47:47.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/17/817417/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"8fa718ae34e149a04bf038a4921a6c48dffe9ce9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-11-11 05:05:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d48ce5e86f52eaf2d4104a7e67de0e923162f095","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 05:06:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3a512385f892cd0144a10facbd46151b6baa821","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 06:27:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10e901838e544bb39902a62d4b6ae5a5 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/495b268ed16b472aac7f3cd02c272d4c : SUCCESS in 9m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4889ec8e0c04062bc69b0785abef817 : SUCCESS in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53604bb3344c471ba91abab6e5ff3584 : SUCCESS in 13m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/42baabe10763447690f5fa3443f9e607 : SUCCESS in 1h 05m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8245b4adc193478a8eba227e935bcb5d : SUCCESS in 1h 07m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b44e076d26e4bae824bee7f37aca806 : SUCCESS in 6m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bbc631257e424b98a3519dbc72439e4b : SUCCESS in 1h 19m 00s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dd159b56bc304f8a925d98144fabcc84 : SUCCESS in 1h 17m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80887fbbbc0847c3a69a2835a864d877 : FAILURE in 1h 00m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c111013dfd8cb992e8782b3b754435ef42409aa8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-11-11 10:39:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"030861b577b22594a6c2d5a219b1630f8aaee1ce","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-11-11 11:44:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/6fc08b3/index.html : SUCCESS in 1h 04m 13s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/defc2f4/index.html : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":1},{"id":"809f94e833c4a623798891114b9657eb92a709a1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-11 12:03:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817417/1/iscsi/ : SUCCESS in 1h 23m 18s\n- cinder-smb http://cloudbase-ci.com/cinder/817417/1/smb/ : SUCCESS in 55m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c0b441d249554a2c19d5ae55d86a2ef644aaad4c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-11 12:13:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/17/817417/1/check/dsvm-tempest-cisco-zonemanager-job/868263d/ : FAILURE in 31m 16s","accounts_in_message":[],"_revision_number":1},{"id":"7474026ca6504a24e24cacb5e2b921b32509a89a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-11 12:15:20.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/17/817417/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"c4939182b29b2fc1c59d400dc23b040de711ab15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 13:08:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f0899c25ba543aba1854931914b789a : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d17fdaf733d14f3c8f094ea4e910b6f7 : SUCCESS in 9m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0508cb72aff943e38efb8b9d1401f7aa : SUCCESS in 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77d22f8c31e04064bd563fbce6a41be9 : SUCCESS in 14m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/675ab0ce441049a5a707dc7b1c1bf43e : SUCCESS in 1h 05m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a8c63252940146bca45b6c39520fba9c : SUCCESS in 1h 13m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19cfba65d95a4ae49d90c69af6c6a472 : SUCCESS in 8m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/54fae1afa69049afadccc8c76352ef1e : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6313a50b1dc1482e98079009b45c6307 : SUCCESS in 6m 55s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/2f716a85090b4bd6976fd7dbac4950ee : SUCCESS in 7m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/272b8fc93a1e42aa96250f24fae7b758 : SUCCESS in 34m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/93384573a5b44333a43a5aef9e0a0e55 : SUCCESS in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0c9e7fc923e546bcae1281cc86a4e43b : SUCCESS in 1h 27m 11s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7cb9ae20c3a548e095984b652ad14fb9 : SUCCESS in 1h 24m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f5f4930e28b749f2b19f8c52e93589ae : SUCCESS in 1h 24m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/394d86d54693407985b02444f601b4a3 : SUCCESS in 1h 10m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1b14c7eee2de4369ab87f891df467c25 : FAILURE in 1h 12m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c5572e7d98074860a592f9a6ec0a0e3e : SUCCESS in 2h 24m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8dfac2636f66452baf77b133b62f3f44 : SUCCESS in 55m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2a05fe7b3a4a35f777ae6aae3ed6d7a852cf2de4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 13:09:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b61c33821ffb8e63b63bd5913f6a4bf510776219","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-11 13:20:26.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic tempest-dsvm-full-bionic : NODE_FAILURE in 0s\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/1571ba848373425e9a52ccca7a60b008 : POST_FAILURE in 2h 07m 07s\n- devstack-plugin-ceph-dsvm-bionic-py3 finger://zuul-v3-executor.novalocal/5f9001a70a464c82a98a887a11cb9e2a : POST_FAILURE in 2h 27m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e8b92b8cff3f68ffed30427093dc2e44aaa79e80","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-11 13:20:48.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/17/817417/1/upstream-check/cinder-cDOT-iSCSI/463175a/ : FAILURE in 11m 20s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/17/817417/1/upstream-check/cinder-cDOT-NFS/66199c7/ : FAILURE in 11m 40s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/17/817417/1/upstream-check/cinder-cDOT-FCP/302fa72/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"030d8e0cbc765e3a22e5ffd826745a5898371f55","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-11-11 13:42:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/17/817417/1/upstream-solidfire-check/cinder-solidfire-iSCSI/dc3d162/ : FAILURE in 32m 41s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"398b6efc0b48c6c6c04632ef8c0122ffd9331b17","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-11 14:06:36.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/17/817417/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d484c6ee2f8a491039bd9932beadee8d3b880935","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 14:38:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a063ca3c81394b989b2ed95abd0339c9 : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d731052e66564334bfe8bca819db65c9 : SUCCESS in 10m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f66ea1d353642628fb703ca1bbe80fb : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dafded854e704ddd8dbd5c12fd747d4f : SUCCESS in 13m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/0a14c3308dce4b14acaf6a8a57b68e7b : SUCCESS in 1h 01m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d71d139f1f564240a3e2d01c64c7f8a1 : SUCCESS in 1h 06m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2369eed25acf4a9db924f7f1ff7ab7b8 : SUCCESS in 7m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/af2141b0f401455dbb0fa7a81dda0d70 : SUCCESS in 1h 12m 36s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bf689dc76d11445dad018193a4699fc4 : SUCCESS in 1h 09m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e343de8ca514a5e99c06a229910d645 : SUCCESS in 52m 03s","accounts_in_message":[],"_revision_number":1},{"id":"04817ef55a319154e8ba0468a0f81f3370aab470","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 14:38:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e4d66b13d2fbbf666d6bb0f75e978b980b78accf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 14:41:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8465fe547c0f4be1bc5bef8c4013628d : SUCCESS in 1m 31s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35a81a8d7e7f4320a63b2963efe3f51f : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0c73ac794fc1e6d435692e95117a9d58b447628e","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-11-18 07:38:25.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/17/817417/1/check/dsvm-tempest-TOYOU-ACS5000-driver/28c4c08 : FAILURE in 7m 28s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/17/817417/1/check/fc-test/78a7e3c : ABORTED in 6m 20s","accounts_in_message":[],"_revision_number":1},{"id":"7006e73eff199dfbd420e06ae2c64bd1e4d655ac","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-18 08:47:19.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/17/817417/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d69f309053bb467e7cb0b10928638d6c381a4b83","revisions":{"d69f309053bb467e7cb0b10928638d6c381a4b83":{"kind":"REWORK","_number":1,"created":"2021-11-10 15:01:04.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/17/817417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/817417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/817417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/817417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/817417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/817417/1"}}},"commit":{"parents":[{"commit":"2ef9db8e4550c016fddc3bdffa137a653fd111d2","subject":"Merge \"RBD: use correct stripe unit in clone operation\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ef9db8e4550c016fddc3bdffa137a653fd111d2"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-03-29 12:32:58.000000000","tz":-240},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-11-10 15:01:04.000000000","tz":0},"subject":"RBD: Open RBD images read-only where possible","message":"RBD: Open RBD images read-only where possible\n\nIn cases where we don\u0027t need to modify the image,\nopen rbd images in read-only mode.\n\nCloses-Bug: #1947518\nChange-Id: I8287460b902dd525aa5313861142f5fb8490e60a\n(cherry picked from commit e644e3584b040823f3b3d9d73730b861a3ec413f)\n(cherry picked from commit 5b169aeea76baecf0b6795efd09dfb1c58ff20d9)\n(cherry picked from commit f2fe6cc115da08670e7aab39fc21b354ce55d3cd)\n(cherry picked from commit 5379af08f0a821f6d4cf1571e83680d55e5c17b8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d69f309053bb467e7cb0b10928638d6c381a4b83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d69f309053bb467e7cb0b10928638d6c381a4b83"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
