)]}'
{"id":"openstack%2Fcinder~817514","triplet_id":"openstack%2Fcinder~master~I9729b83be7568928cf962174f98d632a6ecf3d15","project":"openstack/cinder","branch":"master","topic":"bug/1946618","attention_set":{"34106":{"account":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"last_update":"2022-01-03 05:54:46.000000000","reason":"Someone else replied on the change"},"12670":{"account":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"last_update":"2021-11-30 10:07:28.000000000","reason":"Prajakta replied on the change"}},"removed_from_attention_set":{"32036":{"account":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"last_update":"2022-02-09 07:49:03.000000000","reason":"katari manoj kumar replied on the change"}},"hashtags":[],"change_id":"I9729b83be7568928cf962174f98d632a6ecf3d15","subject":"[SVF] Update correct Error message for add/remove volumes operation in  group-update","status":"NEW","created":"2021-11-11 05:46:39.000000000","updated":"2022-02-09 09:03:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":17,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"8ea6addfb7e342de13ea587cae76cad4a63de205","_number":817514,"virtual_id_number":817514,"owner":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-12-31 07:18:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":-2,"date":"2022-01-18 14:32:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-11 05:49:08.000000000","updated_by":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"reviewer":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"state":"REVIEWER"},{"updated":"2021-11-11 06:09:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-11 06:10:51.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-11 06:27:05.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-11 06:51:59.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-11-11 07:00:27.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-11-11 07:00:47.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-11-11 07:02:13.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-11 07:19:32.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-11 07:25:28.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-11 08:31:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-11 08:50:58.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-11-11 09:07:34.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-11 09:44:59.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-11 12:43:09.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-11-11 13:46:31.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-11 14:10:22.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-11 14:24:06.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-11-11 14:48:09.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-11-11 14:50:49.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-11-11 15:02:03.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-11-11 15:14:44.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-11-12 02:48:56.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-13 15:48:20.000000000","updated_by":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"reviewer":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"state":"CC"},{"updated":"2021-11-18 21:41:24.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-11-19 14:10:15.000000000","updated_by":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"reviewer":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"state":"REVIEWER"},{"updated":"2021-12-10 09:53:24.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2021-12-10 10:20:57.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-30 08:23:04.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-12-30 08:28:55.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-31 04:29:53.000000000","updated_by":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"reviewer":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"state":"CC"},{"updated":"2021-12-31 06:30:12.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2022-01-04 16:42:13.000000000","updated_by":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"reviewer":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"state":"CC"},{"updated":"2022-01-15 15:20:27.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-18 14:32:56.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"4afb786d4793262fb842365fdfe5ce2b7bf6e726","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"date":"2021-11-11 05:46:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"259b4da355bf35a46814eb4317ff8277bfe1ba9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 06:09:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f96df01f01504865a27155552f45c085 : SUCCESS in 15m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5263bbd1fac416ca3df22762adf8325 : SUCCESS in 20m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab6a40a4cb369483b106676cda4f82574ab2d70f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-11 06:10:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/14/817514/1/check/iscsi-driver-nimble/7ab4471/7ab4471.html : FAILURE in 22m 34s","accounts_in_message":[],"_revision_number":1},{"id":"5d76452399c6e6cdec8aab846ddbe3208a2fd1e6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-11-11 06:27:05.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/817514/817514/1/check-not-vote/ext-cinder-zuul/6ed5059 : SUCCESS in 39m 12s","accounts_in_message":[],"_revision_number":1},{"id":"20156608ed9706ae8fed37b25c9ca19cb8df009c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-11 06:51:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817514/1/iscsi/ : SUCCESS in 1h 04m 52s\n- cinder-smb http://cloudbase-ci.com/cinder/817514/1/smb/ : SUCCESS in 48m 03s","accounts_in_message":[],"_revision_number":1},{"id":"ec7da9d2ee3494f65bcf4770d1d65d4be7156b56","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-11 07:00:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/14/817514/1/check/dsvm-tempest-cisco-zonemanager-job/73c56a1/ : SUCCESS in 1h 08m 43s","accounts_in_message":[],"_revision_number":1},{"id":"2d223b9e52cba12df52e1ec5b45e1335aad311b2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-11-11 07:00:47.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/14/817514/1/check/DellEMC_Unity_FC/03a0d5f/DellEMC_Unity_FC/143 : SUCCESS in 1h 13m 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/1/check/DellEMC_Unity_iSCSI/946ec1c/DellEMC_Unity_iSCSI/141 : SUCCESS in 21m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"49ac3100a7622e28a284eb89d1f09ad6fc08bbfa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-11 07:02:13.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"acc6969168da10bce87432566ea13240d0c8c7e2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-11 07:07:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/14/817514/1/check/iscsi-driver-primera/6ffa6cc/6ffa6cc.html : SUCCESS in 1h 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8e49b5a692013b3fa9191d51b2725fe16b4dd68a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-11 07:15:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/14/817514/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/db4d8d5/db4d8d5.html : FAILURE in 1h 26m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2b29aa54e9c1d9f3a5427e10eaaf7decd87d746c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-11-11 07:19:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e3a0058100e04f25a006b7661c6d6d6a : SUCCESS in 1h 31m 01s","accounts_in_message":[],"_revision_number":1},{"id":"818f830c521fbf41d10c5fcf06543bb55261178f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-11 07:25:28.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/14/817514/1/check/DellEMC_XtremIO_FC/1482085/DellEMC_XtremIO_FC/106 : SUCCESS in 1h 00m 32s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/1/check/DellEMC_XtremIO_iSCSI/2ac31f9/DellEMC_XtremIO_iSCSI/133 : SUCCESS in 1h 38m 29s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"a7cf6cc928259d81957016e1de58f4ccd0e13458","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-11 07:26:04.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"9f82051492616ea2d504b0e9cf821e2efbf72bb0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-11 07:28:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/14/817514/1/check/iscsi-driver-alletra-6k/948a213/948a213.html : SUCCESS in 1h 39m 50s","accounts_in_message":[],"_revision_number":1},{"id":"3f600f48af65789e7cb26e8f0680b792e19a690f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 08:31:18.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d125164ed66244a6872845d99aa75d4c : FAILURE in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70dc9deb6e95419cbc0c355bb8aa570f : SUCCESS in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c964b57cf83d47a0b5006393cd72b265 : SUCCESS in 9m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e56ad5b110b9454aa4ad74c071d4bcbf : SUCCESS in 12m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/295b279970894b118b5f235e2ff584ae : SUCCESS in 11m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/c4a9097291034801af6aef009839eb2d : SUCCESS in 1h 00m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c53eaefac3564844bda829bb3912a991 : SUCCESS in 1h 05m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bdd9f98d0984409c941d228f6ee0b92a : SUCCESS in 13m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/84838e200b674763820c3007c4f5ec93 : SUCCESS in 6m 27s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b395a808f7b14aa9ba63219f88a940d1 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8b4cded268c74725a9906802261a03fa : SUCCESS in 6m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/077a5101b3e946a1bc374719502ffa2c : SUCCESS in 6m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6988b0666db54d33bac75ae469b5a538 : FAILURE in 32m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6275e4da8db846b39b471c73ace595fe : FAILURE in 4m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/41eb7563ffc443dca3b827a546f3afc8 : SUCCESS in 1h 21m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6fb5393233d741cd900b56dc8a7b9bdc : SUCCESS in 1h 28m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8eb0bf2e18cb40c9b163d814a81f5df5 : SUCCESS in 1h 25m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/96a5a32be17a4e30b125fec0e7326ee5 : SUCCESS in 1h 02m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4435a3f3610d4df2a76555935a57ede6 : SUCCESS in 1h 13m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bf50a1bccf134b939270db51b99d86b9 : FAILURE in 43m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/03e580ddabd7423294be2200bd3dd02b : SUCCESS in 1h 13m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8238f261f29444f186469b3ff08fbea1 : SUCCESS in 2h 37m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afbf6e202ddd4d02bffb643b594f9a2d : SUCCESS in 50m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28cbc62ac1e049db8eb41d95c4a43bcc : POST_FAILURE in 29m 17s","accounts_in_message":[],"_revision_number":1},{"id":"23e7a8ebf0881fa93bb7eb4d829dcb049102d3a6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-11 08:32:38.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"74d644bb5afb88a20b3e5c1b8de8eae3bece7f42","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-11 08:50:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/14/817514/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b6ffd55/ : TIMED_OUT in 3h 03m 10s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/14/817514/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/db92ff6/ : SUCCESS in 1h 44m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5f56ff2042249e24f3617a70b6c67f884876ac31","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-11 09:07:34.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/47e6ef24b86f47d0aec579ff88e88056 : POST_FAILURE in 1h 49m 38s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/4794257ec6bb4a43a1542985af36ebc9 : TIMED_OUT in 3h 07m 14s","accounts_in_message":[],"_revision_number":1},{"id":"89d81401517bf1e5668bb32caa13cf25b20f235c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-11 09:44:59.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-14-817514-1/ns5_nfs : SUCCESS in 3h 57m 19s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-14-817514-1/ns5_iscsi : SUCCESS in 3h 57m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"cd5ed0742af715e7f976d86a1a86a3c0e4dab734","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-11 09:45:48.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"c8ea4ca654efa4d934bdc89ca2da85b4e7403d68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"date":"2021-11-11 10:22:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aa0af0190601f8cba7c94998a10ebf8ce63c18fe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-11 10:23:59.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"958142fb23dccc21c2c384e55f1f8238edb7aedb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 10:54:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bdb4eac19e7540f7b858c95dc930d200 : SUCCESS in 13m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/002f5087164a4adab84b075d1ba2d325 : SUCCESS in 20m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8cfac10caac5a3f95c2987074cd6f5b925dcac9d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-11-11 11:04:15.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/817514/817514/2/check-not-vote/ext-cinder-zuul/66b2f5f : SUCCESS in 39m 45s","accounts_in_message":[],"_revision_number":2},{"id":"003758c5df29ffb38eac67fd852cf007cb4b7f11","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-11 11:28:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817514/2/iscsi/ : SUCCESS in 51m 56s\n- cinder-smb http://cloudbase-ci.com/cinder/817514/2/smb/ : SUCCESS in 1h 05m 31s","accounts_in_message":[],"_revision_number":2},{"id":"6dfd23cde42e7a6712bc5996b492d365a543d225","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-11-11 11:30:50.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/14/817514/2/check/DellEMC_Unity_FC/012555d/DellEMC_Unity_FC/144 : SUCCESS in 1h 07m 45s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/2/check/DellEMC_Unity_iSCSI/b41a040/DellEMC_Unity_iSCSI/142 : SUCCESS in 21m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"cdceb0d78ab43c30b6506a46364663232de01670","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-11 11:32:03.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"43e6d157be7830c965215986082156f0559df881","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-11 11:36:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/14/817514/2/check/dsvm-tempest-cisco-zonemanager-job/ebdaf43/ : SUCCESS in 1h 08m 29s","accounts_in_message":[],"_revision_number":2},{"id":"6a48bdfa24cac18e8b4a927fc4ce2c75cf65f585","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-11 11:40:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/14/817514/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/01b0d20/01b0d20.html : SUCCESS in 1h 15m 09s","accounts_in_message":[],"_revision_number":2},{"id":"5f809cbfbc2f88ac2458655ced1fd41f82710af4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-11 11:43:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/14/817514/2/check/iscsi-driver-primera/e43bca3/e43bca3.html : SUCCESS in 1h 17m 52s","accounts_in_message":[],"_revision_number":2},{"id":"587b22dd66e61982b6f4e706d5fe7932fbd661ee","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-11-11 11:52:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/63c1349379a440a682968b2d5cb1455f : SUCCESS in 1h 27m 35s","accounts_in_message":[],"_revision_number":2},{"id":"c6337d9aebcee81af05beb670070f32610053004","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-11 12:04:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/14/817514/2/check/iscsi-driver-alletra-6k/9e4c8a9/9e4c8a9.html : SUCCESS in 1h 39m 46s","accounts_in_message":[],"_revision_number":2},{"id":"e7b0013e505d3b632fdba76401a879a23cfc1a62","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-11 12:07:14.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/14/817514/2/check/DellEMC_XtremIO_FC/afc71a1/DellEMC_XtremIO_FC/107 : SUCCESS in 1h 07m 12s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/2/check/DellEMC_XtremIO_iSCSI/9e07329/DellEMC_XtremIO_iSCSI/134 : SUCCESS in 1h 44m 08s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"93ceccf76141b11a2d3250cfa56661a79171020e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-11 12:14:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/14/817514/2/check/iscsi-driver-nimble/d992795/d992795.html : SUCCESS in 1h 49m 13s","accounts_in_message":[],"_revision_number":2},{"id":"9265914a3ec9e89cd402188736865f782bbb29ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 12:17:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4650f4112b04f87bb6744adca6b86ab : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffed303c14ce452f870bb1fc23afa1d5 : SUCCESS in 10m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/813e8e53ee9c46dfabdae229de20db8f : SUCCESS in 10m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4487637cb8ad4cf1813511b8f521ee85 : SUCCESS in 11m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e772b17300f4f14a6435e2cef380999 : SUCCESS in 11m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/bdedddec82ea4304acaaa3516c68365e : SUCCESS in 55m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/06121337672449b5874f02a793ccc407 : SUCCESS in 1h 08m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/609dc33a35114371b6672cc31ac74e37 : SUCCESS in 12m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/78f484f17bdd49afa8aa2e010f58ea5d : SUCCESS in 6m 19s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9d38b8d006904569b8dce6c47f53e37a : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/72a2ba7aefaf4969a9a93b276fda8d47 : SUCCESS in 7m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/51f832fc60e94e5a9727c23e696cef9d : SUCCESS in 6m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eb1d4659e2a048b68a6d21aa66c2ccb7 : FAILURE in 29m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ec3238b6b1544a7f926b443ae7d5aa3e : FAILURE in 4m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ecf58fe606ee403d9c8a9c8e8fd6f4ba : SUCCESS in 1h 15m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0d7b4f17fb6b4bbbb180dd9853ffea40 : SUCCESS in 1h 41m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/38880255db0e4760b4a335636344b52c : SUCCESS in 1h 31m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/74f7ad363dcb435da55880d6d396fa7e : SUCCESS in 1h 06m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ac71b5ff6cb947ee8754ee8a56c8c54a : SUCCESS in 1h 09m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/153def50c58b4a14a64b0de7ddf4fec2 : SUCCESS in 39m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c33869b7c4c948e5855cbc6995d08bad : SUCCESS in 1h 15m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9882db3baf48487083c5a5298fbc86bf : SUCCESS in 1h 51m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e06862d0ea114fc79f23d7624dd58671 : SUCCESS in 48m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/409e73781d314bf5941fd7388fca7fbc : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":2},{"id":"d208c2bc47b9c758afab0a2e9e925fb88ca77e64","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-11 12:22:20.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/14/817514/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/3e721a7/ : FAILURE in 1h 58m 05s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/14/817514/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ab08bd6/ : SUCCESS in 55m 45s","accounts_in_message":[],"_revision_number":2},{"id":"b5da2630df628bb5d920e6b184a9d333433382cb","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-11-11 12:43:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/14/817514/2/upstream-solidfire-check/cinder-solidfire-iSCSI/5142c8f/ : FAILURE in 24m 18s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"8b97c316333c77db0102a51a2f81c5ceb201975b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-11 12:48:20.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/2d832a8dc1bd41d2b0b72e4c964c75f2 : POST_FAILURE in 2h 11m 31s","accounts_in_message":[],"_revision_number":2},{"id":"aa6677e119dcd802cc7eafe3650c0f6393b29199","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-11 12:49:06.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"a47a760c8e8c47c5f8fe8959c34d852edc9ea453","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-11-11 13:26:50.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/14/817514/2/check/DellEMC_Unity_FC/122f393/DellEMC_Unity_FC/145 : SUCCESS in 1h 08m 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/2/check/DellEMC_Unity_iSCSI/ea7b73d/DellEMC_Unity_iSCSI/143 : SUCCESS in 21m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"7362d8f8730e66fdc769690d6974a2e8c82b0861","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-11 13:27:44.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"05a2f0876d53bdccf3b891af3de7b63bb076828c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-11 13:46:31.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/14/817514/2/upstream-check/cinder-cDOT-iSCSI/6b6b97e/ : SUCCESS in 1h 27m 56s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/14/817514/2/upstream-check/cinder-cDOT-NFS/c9114dc/ : SUCCESS in 32m 25s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/14/817514/2/upstream-check/cinder-cDOT-FCP/57c9adf/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"e99e0603bbd31d03834a913fd551204e61def814","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-11 13:56:38.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/14/817514/2/check/DellEMC_XtremIO_FC/3e75a7f/DellEMC_XtremIO_FC/108 : SUCCESS in 1h 05m 50s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/2/check/DellEMC_XtremIO_iSCSI/bfeed66/DellEMC_XtremIO_iSCSI/135 : SUCCESS in 1h 38m 38s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"214b5cb757244dda716807e5ecafa34f96dd40d9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-11-11 14:10:22.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/14/817514/2/check/DellEMC_PowerFlex/78d1d78/DellEMC_PowerFlex/76 : SUCCESS in 1h 51m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"491e09052b095521b83796e54bf6eaeff3128985","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-11 14:11:24.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"5e10d77e9a7d8fcd80463d5824db1d082d1e3f3d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-11 14:22:25.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-14-817514-2/ns5_nfs : SUCCESS in 3h 57m 31s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-14-817514-2/ns5_iscsi : SUCCESS in 3h 58m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"57352b174fa8505b73da628d1a9777eaee09ddf8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-11 14:22:49.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"d6a258ea35671309c772ea9c69e2aa98a4aeb00d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-11-11 14:24:06.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/2/check/DellEMC_PowerStore_iSCSI/79bdfab/DellEMC_PowerStore_iSCSI/54 : SUCCESS in 2h 06m 04s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/14/817514/2/check/DellEMC_PowerStore_FC/2ca72e7/DellEMC_PowerStore_FC/61 : SUCCESS in 1h 18m 10s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"770736f2aa7747a9cf078857811ba6c5132c5405","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-11 14:24:16.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"68c4f2b1407ef70839da1498ac25dde30f46d7b2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-11 14:46:02.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"3d0958e3435846e51f0f34d95fc44bb903881d0b","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-11-11 14:48:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI http://os-logs.tristero.net/dell-ci/817514-2/PVME-iSCSI : SUCCESS in 58:56.21\n- PVME-FC http://os-logs.tristero.net/dell-ci/817514-2/PVME-FC : SUCCESS in 1:01:30","accounts_in_message":[],"_revision_number":2},{"id":"b625c65650c0bb6c6505dcb70fd63c81b6e21c08","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-11-11 14:50:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/817514-2/HPMSA-iSCSI : SUCCESS in 1:18:49\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/817514-2/HPMSA-FC : SUCCESS in 45:31.85","accounts_in_message":[],"_revision_number":2},{"id":"76f86ac8c58c2c7ffc3e4ba6916e5f9edf2ea26d","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-11-11 15:02:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/817514-2/Seagate-iSCSI : SUCCESS in 1:28:38\n- Seagate-FC http://os-logs.tristero.net/seagate-ci/817514-2/Seagate-FC : SUCCESS in 47:41.47","accounts_in_message":[],"_revision_number":2},{"id":"8f0f0a2330e5b9dd87c8a029bd6746bfb8376ad5","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-11-11 15:14:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/817514-2/Lenovo-iSCSI : SUCCESS in 1:33:01\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/817514-2/Lenovo-FC : SUCCESS in 55:47.18","accounts_in_message":[],"_revision_number":2},{"id":"16e2566653722d334812c0cfd86b1404c43e0087","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-12 02:48:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/14/817514/2/20211112053327/NEC-FC : SUCCESS in 6h 11m 48s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/14/817514/2/20211112053327/NEC-ISCSI : SUCCESS in 6h 15m 25s","accounts_in_message":[],"_revision_number":2},{"id":"71112715b7c1d795236e5a7929799e67673dced6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2021-11-13 15:48:20.000000000","message":"Patch Set 1:\n\nBuild Successful. ibm-gpfs-dsvm-tempest-full: https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/123/index.html","accounts_in_message":[],"_revision_number":1},{"id":"ad0be47fdc7909f43463a25fd4845d9ec2d7a1b8","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-11-18 21:41:24.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/14/817514/2/check/dsvm-tempest-TOYOU-ACS5000-driver/5d4cd4c : FAILURE in 2h 23m 24s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/14/817514/2/check/fc-test/f953cce : SUCCESS in 2h 06m 16s","accounts_in_message":[],"_revision_number":2},{"id":"f5cbcf40f5296ed572d585915ac3998fa0808999","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-18 22:57:28.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"6b53895cc3ca6b428b9a740b2357b96b773da1fb","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-11-19 14:10:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ca05d504be7c4b17b89f74eac57937f20564a120","author":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"date":"2021-11-30 10:07:28.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"97a3cc52d3b5be19850be69d7c0257e5ef312109","author":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"date":"2021-11-30 10:38:36.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"76ad709bd40ecaf386a110ba47c90ef0988a999e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"date":"2021-11-30 11:24:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e6ca66650506719e4e05ee10a6877543b8239744","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-30 11:39:37.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817514/3/iscsi/ : FAILURE in 13m 49s\n- cinder-smb http://cloudbase-ci.com/cinder/817514/3/smb/ : FAILURE in 13m 45s","accounts_in_message":[],"_revision_number":3},{"id":"23a269c4842c3f30ba1c59db50a5002c9eaba382","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 11:47:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/475aedd35f5b429a868e9d2b376ddd9d : SUCCESS in 19m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3f778e75ce564761b2ae66ddf7964b04 : FAILURE in 16m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"79e44cd93ed9f2d7b360e9ccd1502a1971fd7c72","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-30 13:07:03.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/14/817514/3/check/iscsi-driver-alletra-6k/4f6346e/4f6346e.html : SUCCESS in 1h 40m 11s","accounts_in_message":[],"_revision_number":3},{"id":"08917eb74ff77635f25721f630b9f84cc5c233ad","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-30 13:17:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/14/817514/3/check/dsvm-tempest-cisco-zonemanager-job/8091c1b/ : SUCCESS in 1h 09m 45s","accounts_in_message":[],"_revision_number":3},{"id":"fd04888d09ee0414ff00b529ac48345d8b5c3d65","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-30 13:47:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/14/817514/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/1dba071/ : SUCCESS in 2h 21m 46s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/14/817514/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/2e8aa36/ : SUCCESS in 1h 04m 28s","accounts_in_message":[],"_revision_number":3},{"id":"cfd55c0f0a833710bf268690938a12a5110ef76a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-11-30 13:54:13.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/14/817514/3/check/DellEMC_Unity_FC/796f23a/DellEMC_Unity_FC/216 : SUCCESS in 1h 09m 40s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/3/check/DellEMC_Unity_iSCSI/a4178e4/DellEMC_Unity_iSCSI/218 : SUCCESS in 1h 43m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"5ffb96b83d06658dc389d8d7125cb3512f2125aa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-30 14:01:05.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/14/817514/3/check/iscsi-driver-nimble/6d1e072/6d1e072.html : FAILURE in 1h 10m 30s","accounts_in_message":[],"_revision_number":3},{"id":"fa2082accba6f63edc2214787c9c3dca82c49bb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 14:05:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f9f6b7d3d741d99e8758a020a49a0d : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85e3dc0d5a594e0f9562234be18bd756 : SUCCESS in 10m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44ba6276feea41fcbb92fc3dee3b1f02 : SUCCESS in 11m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6b0c0d070db40b2a652b3fa38922940 : SUCCESS in 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/27fa0948165949a785bf7cbd617f4b2d : SUCCESS in 1h 00m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7e854d97e1e844e5bff70388f33296c0 : SUCCESS in 1h 08m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5cbf7d8901424c5d9624bca0510fc522 : SUCCESS in 12m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3331db1eea3c4863ab555900bc57ce06 : SUCCESS in 5m 51s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/95e093a9292c4fc1b18bff9af4f69149 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3dbbe5a66d84544ba50d83d8df5e870 : SUCCESS in 6m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/32266d5a35d644108b5c54e9e91300c8 : SUCCESS in 6m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2b0abf777d3d4b359ab6902195a1358f : SUCCESS in 43m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c8242496665846b48cefca55e1f69542 : SUCCESS in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a627a475ec0a424fa3ac6021960c449f : TIMED_OUT in 2h 05m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cd4d5bfe397c4823b798c768f4ed94af : SUCCESS in 1h 24m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/979dcf3b1b2b4fadb20ce3faa26d9cd3 : SUCCESS in 1h 20m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/da7ccc1c54e94c3c90ea5d1da809079e : SUCCESS in 1h 03m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e5c3f4aaf3f94673b0740501bac9c87f : SUCCESS in 1h 18m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9ad9e57c88294089870e7c612f66933c : SUCCESS in 45m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/18945facd032400bbc3432eacd48e898 : SUCCESS in 1h 15m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/532efc8f43e34eeb84802f4a3bfced6b : SUCCESS in 2h 33m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0451ca23ed4463ea58c0297ffe145b0 : SUCCESS in 53m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2abe79923a549c1b9e4d4bcb5cf773d : SUCCESS in 48m 09s","accounts_in_message":[],"_revision_number":3},{"id":"71e0da83c6675a5511344683e889fe3345622719","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-11-30 14:42:35.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1bdd1255d699486c8b7805cff78674c5 : FAILURE in 1h 31m 34s","accounts_in_message":[],"_revision_number":3},{"id":"9822368b7bfe84bb80f129e145b815ed78fb8527","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-30 14:46:13.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/14/817514/3/check/tempest-dsvm-full-focal-py3/35baaa0/ : SUCCESS in 2h 27m 24s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/14/817514/3/check/devstack-plugin-ceph-dsvm-focal-py3/fd7cc44/ : TIMED_OUT in 3h 12m 45s","accounts_in_message":[],"_revision_number":3},{"id":"dcf0ecde8cf2abc18814d9dd552f3c5377a32728","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-30 16:36:23.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-14-817514-3/ns5_nfs : SUCCESS in 4h 19m 12s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-14-817514-3/ns5_iscsi : SUCCESS in 4h 24m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"7c198222c59fca5781d500925c7931467ff44a32","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-30 16:37:30.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"195f08790e65936e722b99510106d6ac3277479f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-30 20:14:09.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"76aef7c9e02700e430868bfa0a66de80491fc769","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-30 20:37:18.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/14/817514/3/check/DellEMC_XtremIO_FC/a1a98e4/DellEMC_XtremIO_FC/255 : SUCCESS in 1h 04m 01s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/3/check/DellEMC_XtremIO_iSCSI/394cedd/DellEMC_XtremIO_iSCSI/272 : SUCCESS in 1h 40m 05s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"02bd855b7b932d7b3d0b5e54be6e88aa728b0893","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-30 23:03:06.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"0f329405dadc69d1bf592f8f8d02540113c35059","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-01 00:29:28.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/14/817514/3/check/dsvm-tempest-TOYOU-ACS5000-driver/ba54193 : FAILURE in 2h 11m 57s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/14/817514/3/check/fc-test/70370e5 : SUCCESS in 2h 00m 31s","accounts_in_message":[],"_revision_number":3},{"id":"a5e49ed595655ff05010c36753cebfcec2a4c195","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-12-01 04:46:39.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/14/817514/3/20211201105639/NEC-ISCSI : SUCCESS in 2h 48m 23s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/14/817514/3/20211201105639/NEC-FC : SUCCESS in 2h 49m 56s","accounts_in_message":[],"_revision_number":3},{"id":"c3707c3c6adfc1abbf02af1bfea0b0b03b0df849","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-01 06:43:39.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"b98b960a135384b7c0125706af21d7c38ded8a29","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-12-05 12:40:19.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/817514/817514/3/check-not-vote/ext-cinder-zuul/482ed8e : SUCCESS in 29m 11s","accounts_in_message":[],"_revision_number":3},{"id":"0144a4dc712cc5399e1298a4a4c85182d8c97988","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"date":"2021-12-10 08:40:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"11228488c13ad7157bb34acde7b1ceb956536234","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 09:03:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d4c582c7e4b4db09af9125b0b750bd5 : SUCCESS in 14m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f32907fa91cd48fbbc1a6a2fdaf94d6b : SUCCESS in 17m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a9615241b14ffc54bd538707a57e35a39dca7bcb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-10 09:35:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817514/4/iscsi/ : SUCCESS in 54m 42s\n- cinder-smb http://cloudbase-ci.com/cinder/817514/4/smb/ : SUCCESS in 49m 31s","accounts_in_message":[],"_revision_number":4},{"id":"a8ff40f8522a1ede86576b55d4a004a86c334119","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-10 09:53:24.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/14/817514/4/check/iscsi-driver-nimble/69c8e17/69c8e17.html : FAILURE in 1h 11m 19s","accounts_in_message":[],"_revision_number":4},{"id":"fc157b316b783393be42912c76ee93ba76c706b7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-12-10 09:55:19.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/14/817514/4/check/dsvm-tempest-cisco-zonemanager-job/92aeff5/ : SUCCESS in 1h 09m 01s","accounts_in_message":[],"_revision_number":4},{"id":"9ebfa325cab081f0765e8cab092ee8db00d82965","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-10 10:03:42.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"e8a3a0f727f036b6bbd4bad79ae4fc98652f8e9b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-12-10 10:20:57.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/14/817514/4/ : SUCCESS in 1h 39m 23s","accounts_in_message":[],"_revision_number":4},{"id":"73bc18ca193402c437c17ff78484dd0496c5b9e4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-10 10:23:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/14/817514/4/check/iscsi-driver-alletra-6k/dd718c4/dd718c4.html : SUCCESS in 1h 41m 02s","accounts_in_message":[],"_revision_number":4},{"id":"d9c6d040927f9822b33c144ef4e7785c976e6224","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 10:51:24.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/206e5be1bbd8410495cf46814ae2fe5e : FAILURE in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6e8dc7cfde24ed495eb8b25b8c2fc24 : SUCCESS in 9m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f46692356f5f47d08d3236cf94ff3351 : SUCCESS in 10m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01e45378a8eb46ef92ea2dada33f7685 : SUCCESS in 12m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/cc68421c5f99433bb024cb50110cda12 : SUCCESS in 42m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aa88c638fb9e46baa2cb0080213198dd : SUCCESS in 1h 01m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cf4ab763a8d40da908359fdb493e6cc : SUCCESS in 37m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/80bb66c9d2744dbfbc9894e0b0615b44 : SUCCESS in 13m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e51644de19424008b0cba17c69bc6fea : SUCCESS in 5m 59s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e64e30081334427b856ba2bf338af354 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9d86e12446dc4f629f7e0e4bddedc354 : SUCCESS in 8m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b728648fe77c450b9c90141f1bffd4cd : SUCCESS in 6m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a861ba8d0cb6425386596d1fe951764b : SUCCESS in 31m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/17ea4c917ecf44b995e24d0686e81d64 : SUCCESS in 7m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cbff787cc294455fbadfce7a1023b9a7 : POST_FAILURE in 16m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fda3945a00924c4aa050c31a93537eb3 : POST_FAILURE in 17m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/28e2527c2aad4cd6a41378803016c192 : SUCCESS in 1h 43m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f59160a02a8d496e9794cd96a512f0ab : SUCCESS in 1h 03m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/adf0bfb110034fea82b5d6c10d8779c2 : SUCCESS in 59m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3d52f1ebf7de479ca9c5c2dc0fe697a8 : SUCCESS in 38m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c2d22756a92e45dfbafac19ccd70387c : SUCCESS in 1h 05m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a339ff5ec6784ff09965a3864896b843 : SUCCESS in 2h 08m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/645fb21ef816459ea932c331f92c2ec8 : SUCCESS in 44m 38s","accounts_in_message":[],"_revision_number":4},{"id":"d390115e6804f3f9631a982e01853741a567bdd5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-10 11:24:02.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/14/817514/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f987c49/ : SUCCESS in 1h 22m 49s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/14/817514/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f14489e/ : FAILURE in 2h 42m 27s","accounts_in_message":[],"_revision_number":4},{"id":"c97bd458ba65fddefdd950aa8ba935a7d1f35655","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-12-10 11:40:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/14/817514/4/20211210174031/NEC-ISCSI : SUCCESS in 2h 52m 11s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/14/817514/4/20211210174031/NEC-FC : SUCCESS in 2h 59m 35s","accounts_in_message":[],"_revision_number":4},{"id":"ae113243fc7b1b71f88a4f4fbef90f458e0bb0b9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-10 12:41:06.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-14-817514-4/ns5_nfs : SUCCESS in 3h 59m 59s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-14-817514-4/ns5_iscsi : SUCCESS in 3h 58m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"73e7183e7f3a2646cda475441cfcd29f5027b0c0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-10 14:03:36.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"d8da8a49903913a9fb3d1f9e619f168f70b57f80","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-11 03:23:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fe9753a9700145989bc9e115b42f7d74 : SUCCESS in 1h 30m 15s","accounts_in_message":[],"_revision_number":4},{"id":"d3793e143438dc0612acbf0f7bb99293cd7dc035","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-13 03:25:03.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/14/817514/4/check/dsvm-tempest-TOYOU-ACS5000-driver/53a726e : SUCCESS in 1h 50m 27s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/14/817514/4/check/fc-test/4fa383d : FAILURE in 2h 08m 49s","accounts_in_message":[],"_revision_number":4},{"id":"9de3341cc207bfbd241121474aecc8f0fb915803","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-13 05:38:15.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"801889869363804a73bff942687fbf5cdab48b6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"date":"2021-12-29 06:47:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a78935a0a621617e841be36e52913d13e64a945f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-29 06:57:21.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/14/817514/5/check/DellEMC_XtremIO_FC/8bbfe38/DellEMC_XtremIO_FC/371 : FAILURE in 9m 55s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/5/check/DellEMC_XtremIO_iSCSI/c33258f/DellEMC_XtremIO_iSCSI/385 : FAILURE in 9m 15s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"1ac0f0607b796a2ce559cba63f164ba928c1f0f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"date":"2021-12-29 07:05:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7218e4d34a632b4fddfac87b07579aa37c34ff94","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-29 07:06:44.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"536ae2faa48d63e19cf0eb0c126dd103565dbea7","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-29 07:15:49.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/14/817514/6/check/DellEMC_XtremIO_FC/2f4caa6/DellEMC_XtremIO_FC/372 : FAILURE in 9m 57s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/6/check/DellEMC_XtremIO_iSCSI/fffc267/DellEMC_XtremIO_iSCSI/386 : FAILURE in 9m 06s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"8151e0e81484ef072ccff81e64bf7e7ba835547a","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-29 07:19:01.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/14/817514/6/check/dsvm-tempest-TOYOU-ACS5000-driver/d52ab85 : FAILURE in 9m 45s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/14/817514/6/check/fc-test/44ddb5a : FAILURE in 11m 44s","accounts_in_message":[],"_revision_number":6},{"id":"2a58f973dbbf63027458fedcae945e1562258749","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 07:27:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/550ba9b533e94d1ab375d5eb6b4e413c : SUCCESS in 12m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/73a3e9f3210342edad8fa6ecc1c8f9f8 : SUCCESS in 17m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"992e18f82d68face7a5995780857a71454a6f4e6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-29 07:31:26.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817514/6/iscsi/ : FAILURE in 25m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/817514/6/smb/ : FAILURE in 25m 09s","accounts_in_message":[],"_revision_number":6},{"id":"bdacfb5db44342269e7f6d8be0ea85120b262094","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-29 08:36:08.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/10975ef9fbbd4f47bbe71a8b54895fe9 : FAILURE in 1h 28m 11s","accounts_in_message":[],"_revision_number":6},{"id":"ca25e96c4460b0d679d727a7c8e183ab5f73370f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 10:13:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4153221de794a3082e91c9246f1edd0 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/781fa90d59a8488b8edb2a6d3435d7ea : SUCCESS in 11m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f991341580e8452bbba4de94f7c6faf8 : SUCCESS in 10m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/882905df08834b13a70574dc38f7befb : SUCCESS in 12m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/a27ef117df7b42c1bbdbdb53bf5cd3f8 : FAILURE in 38m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/79f538122340415ba94513ecc3bac3df : FAILURE in 1h 33m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b393ca21254144539860d43c525c007f : SUCCESS in 38m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5c3044cc18e347dab23e4d0cfb589c36 : SUCCESS in 11m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/30455a033b424e379c11cfe3578c5bea : SUCCESS in 5m 45s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/653dfde6946a4a81bec56dad93c9e532 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5529857b8ce94e43bbd11ed08094b9d1 : SUCCESS in 5m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3cee517e8a684b01ae3b9c3b7ca23d48 : SUCCESS in 6m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/91b1c26a417d4334aa594deb5605c62a : SUCCESS in 44m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/da0d8336db6e4c459cee8b81b7ede6c0 : SUCCESS in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/54da4b3994744fc9bd746c507f2e5487 : SUCCESS in 1h 06m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6c59b05d563f46139375f13ce36a7a5d : SUCCESS in 1h 15m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7ba28e9fe1b24535809eff8ae53860ee : FAILURE in 1h 53m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/84bd23656e854b4db19999d364de658a : FAILURE in 47m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/34b32990c618402bb40899118eeca200 : FAILURE in 1h 49m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/12d201bf1d16435a8b7c6d4a610fadc7 : FAILURE in 38m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0048901da38b4e6eb93370760da03b86 : FAILURE in 1h 27m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/36eee63ce81742c3a624f28aa3041034 : TIMED_OUT in 3h 05m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93e20c53955b489daf2513f9dabd7f61 : FAILURE in 55m 43s","accounts_in_message":[],"_revision_number":6},{"id":"986ac94f40067f7532a9aca856f6b50481eee4e7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-29 10:20:44.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/14/817514/6/check/tempest-dsvm-full-focal-py3/36160ed/ : SUCCESS in 1h 56m 50s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/14/817514/6/check/devstack-plugin-ceph-dsvm-focal-py3/2acb18e/ : TIMED_OUT in 3h 10m 51s","accounts_in_message":[],"_revision_number":6},{"id":"572b0de5b8b1b6eaeebc0be47a966414f9066f25","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-29 10:21:38.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/6/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"9dd9620018e6e2cb21ef83f9e838885f7130b6d2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-29 12:51:17.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/6/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"333f13fc86b51814429cee5cfbf58875a87a18de","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-29 13:30:03.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-14-817514-5/ns5_nfs : FAILURE in 6h 42m 19s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-14-817514-5/ns5_iscsi : FAILURE in 6h 40m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"d1957ff6e30b4a0d02f16c289c0daa7d9fffc616","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-29 14:20:04.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-14-817514-6/ns5_nfs : FAILURE in 6h 44m 9s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-14-817514-6/ns5_iscsi : FAILURE in 6h 47m 18s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"a5234716dd5d13b470c181c7073340d75f617bf5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-29 14:20:56.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"080faaccd08a301e4f15672963bb7cbecd25e17c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-29 15:17:33.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/3fc767f60f8a47cba14a2296b14251da : RETRY_LIMIT in 2h 51m 57s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/78bb481eafaa4b328f5193d387ca01e3 : RETRY_LIMIT in 2h 09m 24s","accounts_in_message":[],"_revision_number":6},{"id":"18102fe4b4551afc0e98174ff81830874d28617d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-29 16:47:35.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/6/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"d9a9544923ebea4f43b717d50a41b30c9915f629","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-12-29 17:18:47.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/14/817514/6/20211229192238/NEC-FC : FAILURE in 6h 42m 41s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/14/817514/6/20211229192238/NEC-ISCSI : FAILURE in 6h 56m 05s","accounts_in_message":[],"_revision_number":6},{"id":"c408173255969d6c826dee0134f8845250ee97a2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-29 18:36:48.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/6/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"f16f0c300dcd7b1d542d8a97399fffe9b6ca26ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"date":"2021-12-30 06:00:26.000000000","message":"Uploaded patch set 7: New patch set was added with same tree, parent, and commit message as Patch Set 6.","accounts_in_message":[],"_revision_number":7},{"id":"72c97cb8a7ff1cb53ac54c28a8ecc618c6241b26","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-30 06:10:20.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/14/817514/7/check/DellEMC_XtremIO_FC/fdfbfa8/DellEMC_XtremIO_FC/374 : FAILURE in 9m 33s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/7/check/DellEMC_XtremIO_iSCSI/9b1404c/DellEMC_XtremIO_iSCSI/388 : FAILURE in 9m 04s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"6c79d37918985d96b4879f5fabaa7f93262f7a8e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-30 06:10:43.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/7/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"bc136dcb67e9dccb1d3abf5d0cf07b54d047d56c","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-30 06:16:46.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/14/817514/7/check/dsvm-tempest-TOYOU-ACS5000-driver/316ce95 : FAILURE in 13m 29s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/14/817514/7/check/fc-test/f9318ce : FAILURE in 15m 08s","accounts_in_message":[],"_revision_number":7},{"id":"7b10b32e71feae53b2e67f77b1d35dcb4a4186e1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-30 06:17:09.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/7/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"4b029c759a92e433d36fd95cf73f33507ef44227","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 06:20:30.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f2926260ec04d6ba26e7227aabcdd0c : SUCCESS in 17m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bbbeb59be3004f5b875f7a2fc5bded41 : SUCCESS in 18m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"469d5bb0ba780a1e3f727047fb175df9fa902dd2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-30 06:22:15.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817514/7/iscsi/ : FAILURE in 21m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/817514/7/smb/ : FAILURE in 21m 19s","accounts_in_message":[],"_revision_number":7},{"id":"bb342ee6b9fa4629654bd67e05c11d81127c370c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-30 06:22:37.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/7/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"a3785f42363c53bb78ed663f6266bd0ee502fdc9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-30 07:24:23.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b30f8cb50f6c4f73801fd00957b70943 : FAILURE in 1h 22m 09s","accounts_in_message":[],"_revision_number":7},{"id":"ea4b7f61c219f40bf18cc6c2fc1060931fab7ffa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-30 07:32:05.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/7/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"a0f6db84c3bdc9a3c26d029b482d29cb56cd3ed3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 08:13:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef4ee0bc02f845b78aad6b507a0420d7 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce73d7cd1c8e4f049c7c40bde18e3f3b : SUCCESS in 9m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63cb5868358d444b8ba6bd707fd06386 : SUCCESS in 12m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c38633ccde3745a08ffe3061f1b43ce8 : SUCCESS in 12m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/a9f4281d0f6d4802aa929fccfa6c6f2b : SUCCESS in 53m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7b1692ee4c764bb4ac923d12bca7fbaf : SUCCESS in 1h 13m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14f7ffa3353c42349fe89b880b535cdd : SUCCESS in 37m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a43985d849364603818554a10280f216 : SUCCESS in 12m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/98c4f0da9bac4764af65c46281b48ade : SUCCESS in 8m 06s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2e161b63df834ad0bb772676516cd94e : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/133bbacd20ce4e629cb3da8b659b2546 : SUCCESS in 9m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e365c63e2d2c4c97bf49a4db18584527 : SUCCESS in 7m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e38dbf73f8f0496eaa232d8bbb45e067 : SUCCESS in 31m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2ed5a8f27600467783da878c1d0ded28 : SUCCESS in 6m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/512ee5cba10d4168866287fc85037435 : SUCCESS in 1h 09m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a94961551a6942dba07f56e719950187 : SUCCESS in 1h 39m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6296a2aaf2d246fea5fd59c8c172ce0f : SUCCESS in 1h 30m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/32c6b5379edb4aedb0fc4366a91d6316 : SUCCESS in 1h 03m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/da018c530ffa469aad08cbd2dbe6e097 : SUCCESS in 1h 09m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4ef0f8de1649422b9d31a729ec111b51 : SUCCESS in 38m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/391eaf9b44df42b8a64586cdb2c72645 : SUCCESS in 1h 10m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e940993024a54a42bc011ffa20d9630f : SUCCESS in 2h 10m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52cdfb70bd814ccba10a718fc03e040d : SUCCESS in 49m 10s","accounts_in_message":[],"_revision_number":7},{"id":"16ab142f2c3068815dbe53f3f5e490edbb56900c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-12-30 08:23:04.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/14/817514/7/check/DellEMC_SC_FC/88b9c61/DellEMC_SC_FC/150 : FAILURE in 8m 08s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/7/check/DellEMC_SC_iSCSI/f8083c4/DellEMC_SC_iSCSI/148 : FAILURE in 8m 50s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"1458600fa72bbd99d626792c7f605fa1605f3d16","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-30 08:23:46.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/14/817514/7/check/DellEMC_XtremIO_FC/21ac18f/DellEMC_XtremIO_FC/376 : FAILURE in 9m 30s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/7/check/DellEMC_XtremIO_iSCSI/166542e/DellEMC_XtremIO_iSCSI/390 : FAILURE in 8m 56s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"d7e3079ca3629254553928cb7d9baba3babf9474","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-12-30 08:28:55.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/7/check/DellEMC_VNX_iSCSI/ca28025/DellEMC_VNX_iSCSI/175 : FAILURE in 9m 47s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/14/817514/7/check/DellEMC_VNX_FC/f9153b9/DellEMC_VNX_FC/182 : FAILURE in 14m 43s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"15286a6f272d4624035afa39e1397c5d97729f67","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-12-30 09:09:25.000000000","message":"Patch Set 7:\n\nBuild failed.\n- Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/817514-7/Seagate-iSCSI : FAILURE in 12:36.73\n- Seagate-FC http://os-logs.tristero.net/seagate-ci/817514-7/Seagate-FC : FAILURE in 12:53.01","accounts_in_message":[],"_revision_number":7},{"id":"3e7bed8e6fb94d93c30d4334961bcdb9d469750a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-30 09:17:28.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/14/817514/7/check/tempest-dsvm-full-focal-py3/8963786/ : TIMED_OUT in 3h 09m 51s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/14/817514/7/check/devstack-plugin-ceph-dsvm-focal-py3/23ae6e4/ : TIMED_OUT in 3h 08m 49s","accounts_in_message":[],"_revision_number":7},{"id":"6533bebee84e87581d5f56ad3556e2eccb20068f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-30 10:09:46.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-14-817514-7/ns5_nfs : SUCCESS in 4h 7m 37s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-14-817514-7/ns5_iscsi : SUCCESS in 4h 8m 44s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"57f7becd8041b475c961b9968e65117091fdfe1d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-12-30 10:11:07.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/14/817514/7/upstream-check/cinder-cDOT-iSCSI/21da715/ : SUCCESS in 1h 56m 15s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/14/817514/7/upstream-check/cinder-cDOT-NFS/ee499a0/ : FAILURE in 1h 28m 22s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/14/817514/7/upstream-check/cinder-cDOT-FCP/616d5b0/ : SUCCESS in 1h 15m 33s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"14b20dce6cbb0e37e091e41ca1fc1d8f54667c28","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-12-30 10:50:35.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/14/817514/7/upstream-solidfire-check/cinder-solidfire-iSCSI/224e033/ : SUCCESS in 2h 35m 25s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"d4a84595cfa76787743177a3c7bd82ee6e8b979c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-30 11:40:28.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/476af7259d114e30a9afa26b134cd8d2 : RETRY_LIMIT in 1h 35m 15s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/a5be6bd952e245c5909ccb734033f7c5 : RETRY_LIMIT in 1h 10m 52s","accounts_in_message":[],"_revision_number":7},{"id":"25c89db4a5eb6da597efdbd2767442af53e7f23b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"date":"2021-12-30 12:04:54.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7f9c49e733fe84ce29b28f98a471b793c480b7ac","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-30 12:05:52.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/7/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"844af5e86f808568440f11184fc5378f3ceb0afc","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-30 12:14:43.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/14/817514/8/check/DellEMC_XtremIO_FC/97c3982/DellEMC_XtremIO_FC/381 : FAILURE in 9m 30s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/8/check/DellEMC_XtremIO_iSCSI/0eedfab/DellEMC_XtremIO_iSCSI/395 : FAILURE in 9m 07s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"d1618db7fc82c79b0a80aa8009c252bea61e7bd7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-30 12:26:57.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817514/8/iscsi/ : FAILURE in 21m 26s\n- cinder-smb http://cloudbase-ci.com/cinder/817514/8/smb/ : FAILURE in 21m 30s","accounts_in_message":[],"_revision_number":8},{"id":"c329b96198466957f9e9efd7c58cd680b426d3b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 12:27:54.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0b468e609f1431baaf5af73d9be5623 : SUCCESS in 16m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff00a9e8c46b4d458dd3571b76645ee4 : SUCCESS in 18m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bcbe83c6e9a6326c651d99cb94a43caeb46e61f9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-30 13:30:13.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"4b211f7795804cb656b04790ec3d9a957a08e8d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 14:19:37.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/351cb772dec14691bb8a035c12771c41 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60b3a5a03c4d4885879462f083dd24cd : FAILURE in 9m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5522840ff09b4327889a5b2d614cd820 : SUCCESS in 10m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ecc7b1ab3c24687976f92530aad0eff : SUCCESS in 13m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/e9ae08f021b34c29abbcb2b505ec06f5 : SUCCESS in 1h 00m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/433c57cffd244f22bb4bdf7839c9550c : SUCCESS in 1h 11m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/319b41128b0d4475b1571b96eec6c77c : SUCCESS in 48m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/427dc996ac3f4aa98c099eeee5d34275 : SUCCESS in 7m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f4910f69511b41ce94f20ad5a441c458 : SUCCESS in 16m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c476308ae23d4841bab699feed3042dc : SUCCESS in 6m 15s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/aa939c4339624e02b0d9114689705125 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c59ee682498a4ca4b930a03b6c21049f : SUCCESS in 7m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8b2adfbdbd7f4d1d9cbffe1c2cc306db : SUCCESS in 6m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/da013d7641df488393a133d9bd9883ce : SUCCESS in 42m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e78c61518e844c23bf552e6bd69e8e8a : SUCCESS in 6m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e2bd8e7234ce4672a701a4bbec26561b : SUCCESS in 1h 16m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b5b6580238d64ef986a7423ee432b53d : SUCCESS in 1h 20m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2bacb0761bc943858493164c4122764b : SUCCESS in 1h 48m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2c213aad5e8a4db7ba71135f40bf77b5 : SUCCESS in 1h 04m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9febeb91b49146ca9aa716d0408b6d97 : SUCCESS in 1h 32m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e12c4ebc0b0d4749b1e9925d81c39395 : SUCCESS in 38m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/70b016b8ff7448768b64e61f36657b1c : SUCCESS in 1h 45m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/84107ddf4e014a00ab058e345b0dc373 : SUCCESS in 2h 09m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de6d76f918524d4489e4c882e4d7b713 : SUCCESS in 1h 19m 42s","accounts_in_message":[],"_revision_number":8},{"id":"c14c0cc58f0374d3541ff97e1e0bf828acc07a8a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-30 14:34:17.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c2c76286c19a4c65bd65229b797dc34d : FAILURE in 1h 25m 09s","accounts_in_message":[],"_revision_number":8},{"id":"e8113b1ac85f2e6dae0bfe9ff7d16b30913697c2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-30 15:02:10.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/14/817514/8/check/tempest-dsvm-full-focal-py3/631a624/ : SUCCESS in 2h 16m 55s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/14/817514/8/check/devstack-plugin-ceph-dsvm-focal-py3/22476e6/ : SUCCESS in 2h 43m 13s","accounts_in_message":[],"_revision_number":8},{"id":"9bce939a958bf69fe68c75fb19214fd37c853c62","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-30 15:57:33.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"7dbc4f344a349fe3d2057375cda3e92813d770d0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-30 16:29:38.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-14-817514-8/ns5_nfs : SUCCESS in 4h 15m 20s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-14-817514-8/ns5_iscsi : SUCCESS in 4h 18m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"b62ed427668b3128143d70227385056a55aa4c22","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-30 17:24:59.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"73fcf23aa6b6b908d3380a73550833d15d83b62b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-30 20:19:16.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/e1d39a40b7284f60b805f6bd73ea2037 : RETRY_LIMIT in 2h 51m 15s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/ad8a31953fb949c98ad4c93df440f682 : RETRY_LIMIT in 1h 05m 05s","accounts_in_message":[],"_revision_number":8},{"id":"cb87901082be44fe3a0e3d19f7d88c7b6c8c1c8d","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2021-12-31 04:29:53.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d4d2f4230307541bd631e8f7a10fc20a91a96bb3","author":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"date":"2021-12-31 05:13:15.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"75719ac8e3ecc66377f926c3dfa84d02542414cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-31 05:35:06.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3cd7c3da0ae8416e8870f458e063d129 : SUCCESS in 18m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15775482ef514dd0aa79aed5a2d77384 : SUCCESS in 16m 38s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fe8be5200dbac7b1322f63324d393994549a40cc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-31 05:37:28.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/817514/8/iscsi/ : FAILURE in 22m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/817514/8/smb/ : FAILURE in 23m 47s","accounts_in_message":[],"_revision_number":8},{"id":"2a49266108b6199991df290865f79d4f9bf20741","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-31 05:38:06.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"c17b032512f48f9a3aa2e76fadf71ac54a5d4f6b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-31 05:45:51.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/14/817514/8/check/DellEMC_XtremIO_FC/e630b2a/DellEMC_XtremIO_FC/386 : SUCCESS in 50m 21s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/8/check/DellEMC_XtremIO_iSCSI/bbe2e80/DellEMC_XtremIO_iSCSI/400 : SUCCESS in 1h 15m 41s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"2735cc08147af925a5f5072e6b650ece1dd203ab","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-12-31 06:30:12.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/e86667d/index.html : FAILURE in 1h 16m 27s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1309a51/index.html : SUCCESS in 1h 05m 52s","accounts_in_message":[],"_revision_number":8},{"id":"79f377a6e30cc65a12991019dd112bd39fe1a811","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-31 06:30:35.000000000","message":"Patch Set 8:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/14/817514/8/check/dsvm-tempest-TOYOU-ACS5000-driver/5ff8165 : FAILURE in 8m 01s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/14/817514/8/check/fc-test/e41cd6e : FAILURE in 11m 07s","accounts_in_message":[],"_revision_number":8},{"id":"5c1f85306f8b59de560d60c2687a5c154a28db26","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-31 06:36:23.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b80de4e25b844464bc9719e9133b1e84 : FAILURE in 1h 21m 40s","accounts_in_message":[],"_revision_number":8},{"id":"4b374811b9babc8ceb10ace736c222d1e0b61adb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-31 07:18:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f9df43abef545f1bdc92d944bb4fe17 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba67f69bfeb7451e8d727ac61f9a979e : SUCCESS in 10m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/187206db2048459789dff675d7e0069e : SUCCESS in 10m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b4f7f2307874575b20f07d8f4a943af : SUCCESS in 14m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/e2babb6523a946778f155f678e8f7d2a : SUCCESS in 58m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d793fdd5f36c45eca70d1a904c3f5cab : SUCCESS in 1h 05m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8f9c63d3f504fbab845ce23aee55a8c : SUCCESS in 49m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/179d7e82bddc47f0afb19d150895b515 : SUCCESS in 6m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a48bcc4466114ecfae4c1679bb48f86a : SUCCESS in 13m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7f187aef7a864f5a912e64245870c724 : SUCCESS in 7m 35s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/98fc90507b0e4dff9296a49192e64cf7 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/32d0af5333634025b2ae72c102092636 : SUCCESS in 8m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e78d3cdc4b6945cb8eef49f255a7cf3b : SUCCESS in 8m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9582c7f143644767a6632c16c2021dcc : SUCCESS in 36m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6f2f93c87e524173875c28e94168b19b : SUCCESS in 7m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a5c72fef9e9641c4ae55074c2223e7aa : SUCCESS in 1h 05m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/187c72a06b3844758b83b52773aed177 : SUCCESS in 1h 14m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cb209a064e404d0ea165305477d3048a : SUCCESS in 1h 18m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/70186a59b00a4cca9729b17a3429a080 : SUCCESS in 57m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a512ec0d96e94659bd1ddf54173c6060 : SUCCESS in 1h 04m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/526dbee47dba4c04b39399240a2dd423 : SUCCESS in 37m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d4b15c4b8b8349babdb967748953b4aa : SUCCESS in 1h 01m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cecefd40ce34495d92fff8e4999ec83e : SUCCESS in 2h 01m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bdcb22b010494b2790d2e7420a53fa82 : SUCCESS in 49m 07s","accounts_in_message":[],"_revision_number":8},{"id":"aaf1fb983898cfc3757fa74a6766f9590895aa49","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-31 07:24:19.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"49f213232f7fa75244a486704b30995bf2dfd514","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-12-31 07:37:29.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/14/817514/8/upstream-solidfire-check/cinder-solidfire-iSCSI/58fae76/ : FAILURE in 18m 12s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"d833670a68362c820b4ffd56897807551f69a220","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-31 07:37:52.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/14/817514/8/check/tempest-dsvm-full-focal-py3/d0ca272/ : SUCCESS in 1h 50m 14s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/14/817514/8/check/devstack-plugin-ceph-dsvm-focal-py3/6f82ee3/ : SUCCESS in 2h 20m 10s","accounts_in_message":[],"_revision_number":8},{"id":"d8b5d131c4f3e90e7290a54a65cd98fa392fafbe","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-12-31 08:12:07.000000000","message":"Patch Set 8:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/817514-8/HPMSA-iSCSI : FAILURE in 12:24.23\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/817514-8/HPMSA-FC : FAILURE in 12:22.13","accounts_in_message":[],"_revision_number":8},{"id":"fa315ea6ee59036ddde0eb96a00e1742ca012e17","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-12-31 08:14:06.000000000","message":"Patch Set 8:\n\nBuild failed.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/817514-8/Seagate-iSCSI : FAILURE in 12:54.29\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/817514-8/Seagate-FC : FAILURE in 12:51.46","accounts_in_message":[],"_revision_number":8},{"id":"63492d704f1619786362be0fae6429d8f04eea55","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-12-31 08:32:22.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/14/817514/8/upstream-check/cinder-cDOT-iSCSI/a985fa9/ : FAILURE in 21m 25s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/14/817514/8/upstream-check/cinder-cDOT-NFS/91fdc7a/ : SUCCESS in 1h 13m 22s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/14/817514/8/upstream-check/cinder-cDOT-FCP/d57dfc7/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"5351b084f81a1ec09779a4b0aac20fe3104d0a41","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-31 08:32:35.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/14/817514/8/check/DellEMC_Unity_FC/9019c01/DellEMC_Unity_FC/303 : SUCCESS in 1h 06m 15s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/8/check/DellEMC_Unity_iSCSI/8717635/DellEMC_Unity_iSCSI/301 : SUCCESS in 1h 14m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"8d56eb137d35d0a170be42fcbce98d1eaf7c6a4d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-31 08:32:46.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"09166e5d0c5882691ff58857fbb826b6fb144a16","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-12-31 08:36:07.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/14/817514/8/check/DellEMC_PowerFlex/0621c25/DellEMC_PowerFlex/219 : SUCCESS in 1h 17m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"e4bcfeb38d3096d730efcdd3101a26a85d67c5ff","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-31 08:36:18.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"ea8142e307a9a95b897f6de3286abf1e8aebd41e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-31 08:48:08.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/14/817514/8/check/DellEMC_XtremIO_FC/5991f3d/DellEMC_XtremIO_FC/387 : SUCCESS in 1h 05m 17s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/8/check/DellEMC_XtremIO_iSCSI/dd4cb6f/DellEMC_XtremIO_iSCSI/401 : SUCCESS in 1h 29m 47s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"278f288340aede6d1bad1aefd79f7740c324b856","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-31 08:48:45.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"9a11d418c4829df0a01ad97796ca8d0e3d0fc36d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-12-31 08:51:18.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/8/check/DellEMC_PowerStore_iSCSI/6403783/DellEMC_PowerStore_iSCSI/191 : FAILURE in 1h 32m 54s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/14/817514/8/check/DellEMC_PowerStore_FC/76184d7/DellEMC_PowerStore_FC/195 : SUCCESS in 56m 41s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"d74cfa5edf083fd763672872e09237eb34d17b6d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-31 08:51:28.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"d9b470c14b9c0cd0b5ac8ac8f5aa8a50af966d7e","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-12-31 09:01:41.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/14/817514/8/check/DellEMC_SC_FC/3c09994/DellEMC_SC_FC/158 : SUCCESS in 48m 16s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/8/check/DellEMC_SC_iSCSI/2980858/DellEMC_SC_iSCSI/155 : SUCCESS in 1h 43m 20s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":8},{"id":"02e9e7f39825d7b7bf7cb45a5bf88c7640666df6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-31 09:02:06.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"525ca7a22dbfaeeb4a5d3aa35180f1877ade4b70","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-12-31 09:33:42.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/8/check/DellEMC_VNX_iSCSI/13c4fe8/DellEMC_VNX_iSCSI/182 : SUCCESS in 2h 12m 19s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/14/817514/8/check/DellEMC_VNX_FC/43f7226/DellEMC_VNX_FC/189 : FAILURE in 15m 17s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"7c5f8f820c0c2b1c13c4ce4ba7a2682566ebcec2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-31 09:34:28.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"4b782edcd84e8555a305ef3c1bea009dd0c1510f","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-12-31 09:52:16.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/817514-8/PVME-iSCSI : SUCCESS in 1:03:02\n- PVME-FC https://os-logs.seagate.com/dell-ci/817514-8/PVME-FC : SUCCESS in 59:15.27","accounts_in_message":[],"_revision_number":8},{"id":"ca94d033a4993a9ed01a34a0bae1b36489155744","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-31 10:24:50.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"f322e51b5ddd30e8e1790b39e442059edeb0cab3","author":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"date":"2022-01-03 04:48:37.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"690ae48b8012e02ec9e1271118fe3a11e21c9b34","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-01-03 05:54:46.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/b12d941/index.html : SUCCESS in 1h 05m 33s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/2f6fa64/index.html : SUCCESS in 1h 03m 32s","accounts_in_message":[],"_revision_number":8},{"id":"9a17283beef38f44acf86e20b3e11dc74978c5fb","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2022-01-04 16:42:13.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"abd3f20af14c12ac777452ee206bbd8537121991","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-01-04 20:08:07.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/14/817514/8/20220104230729/NEC-FC : SUCCESS in 2h 44m 23s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/14/817514/8/20220104230729/NEC-ISCSI : SUCCESS in 6h 00m 35s","accounts_in_message":[],"_revision_number":8},{"id":"a80d70fdad6154cce639df61cad42ae7cb7f7df1","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2022-01-04 21:31:37.000000000","message":"Patch Set 8:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/14/817514/8/check/dsvm-tempest-TOYOU-ACS5000-driver/d36235a : SUCCESS in 2h 46m 58s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/14/817514/8/check/fc-test/7eac23f : SUCCESS in 1h 56m 25s","accounts_in_message":[],"_revision_number":8},{"id":"066791859bfa243914c3fa663f394bc3cf38088f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-04 22:43:37.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"4ea4b712621682ee209f37362079c27eec276cb7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-05 02:33:38.000000000","message":"Patch Set 8:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":8},{"id":"7fbce84ab0f531c1a75917249c6e70d29356010e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-05 02:33:42.000000000","message":"Patch Set 5:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":5},{"id":"9ca811021a21f5619e69fc432ed33138e51580e2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-05 04:04:22.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/817514/817514/8/check-not-vote/ext-cinder-zuul/c86b32c : FAILURE in 30m 22s","accounts_in_message":[],"_revision_number":8},{"id":"017bf0e91e50b3e2b100c29492f9be3b474e41f4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-05 05:22:14.000000000","message":"Patch Set 7:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":7},{"id":"6cdfd83bd98fd2220a31bfb1d83f5d99dfc37a4f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-05 05:22:20.000000000","message":"Patch Set 6:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":6},{"id":"7475be66d706fe79fc73cdee356d55c352891eba","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-01-15 15:20:27.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/14/817514/8/check/DellEMC_PowerMAX_FC/d339cea/DellEMC_PowerMAX_FC/144 : FAILURE in 1h 24m 06s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/14/817514/8/check/DellEMC_PowerMAX_iSCSI/dca51dc/DellEMC_PowerMAX_iSCSI/151 : SUCCESS in 2h 48m 48s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"d5d86d8b1b7d5e64090776bb40ca2bb48369b2c7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-15 16:13:19.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/14/817514/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"fbca3d4e26b79877e4ee0db025d34a6584f0f66d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-01-18 14:32:56.000000000","message":"Patch Set 8: Code-Review-2\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"aeaa5c150e9737d6bd694b9bd0b3d553cd4fb395","author":{"_account_id":32036,"name":"katari manoj kumar","email":"katkumar@in.ibm.com","username":"katarimanojkumar"},"date":"2022-02-09 07:49:03.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8ea6addfb7e342de13ea587cae76cad4a63de205","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-02-09 09:03:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/6c8e366/index.html : SUCCESS in 1h 13m 10s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/1709bae/index.html : SUCCESS in 1h 09m 33s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"23665c2119a73d8b54e0b52109607977c83c90e7","revisions":{"a9adb7ac3489b59aa07a69685857e180a2b83586":{"kind":"REWORK","_number":1,"created":"2021-11-11 05:46:39.000000000","uploader":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"ref":"refs/changes/14/817514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/817514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/817514/1"}}},"commit":{"parents":[{"commit":"3e048975d0f75b983c69974945f8e5a9792c488a","subject":"Merge \"Correct type annotations for utils.retry()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e048975d0f75b983c69974945f8e5a9792c488a"}]}],"author":{"name":"PrajaktaB","email":"prajakta.ambulkar@ibm.com","date":"2021-11-11 05:38:54.000000000","tz":0},"committer":{"name":"PrajaktaB","email":"prajakta.ambulkar@ibm.com","date":"2021-11-11 05:38:54.000000000","tz":0},"subject":"[SVF] Update correct Error message for add/remove volumes in  group-update","message":"[SVF] Update correct Error message for add/remove volumes in  group-update\n\n[Spectrum Virtualize Family] Add/Remove volumes (group-update) from the group does not show proper error or log messages.\nWhen list of volumes are given to add them to the group, if a volume is already exists in the group\nthen it just silently removes the volumes from the list to be added (add_volumes_list) to the group without giving an error/warning mesaage.\nUpdating the cinder code to  print a message in logs that the given volume already exists hence\nthe add volume operation for this volume is not required .\n\nCloses-Bug: #1946618\nChange-Id: I9729b83be7568928cf962174f98d632a6ecf3d15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a9adb7ac3489b59aa07a69685857e180a2b83586"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a9adb7ac3489b59aa07a69685857e180a2b83586"}]},"branch":"refs/heads/master"},"bdab8f8b93786bbb15e0a96918662bf52da6c219":{"kind":"REWORK","_number":2,"created":"2021-11-11 10:22:47.000000000","uploader":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"ref":"refs/changes/14/817514/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/817514/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/817514/2"}}},"commit":{"parents":[{"commit":"3e048975d0f75b983c69974945f8e5a9792c488a","subject":"Merge \"Correct type annotations for utils.retry()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e048975d0f75b983c69974945f8e5a9792c488a"}]}],"author":{"name":"PrajaktaB","email":"prajakta.ambulkar@ibm.com","date":"2021-11-11 05:38:54.000000000","tz":0},"committer":{"name":"PrajaktaB","email":"prajakta.ambulkar@ibm.com","date":"2021-11-11 10:21:14.000000000","tz":0},"subject":"[SVF] Update correct Error message for add/remove volumes in  group-update","message":"[SVF] Update correct Error message for add/remove volumes in  group-update\n\n[Spectrum Virtualize Family] Add/Remove volumes (group-update) from the group does not show proper error or log messages.\nWhen list of volumes are given to add them to the group, if a volume is already exists in the group\nthen it just silently removes the volumes from the list to be added (add_volumes_list) to the group without giving an error/warning mesaage.\nUpdating the cinder code to  print a message in logs that the given volume already exists hence\nthe add volume operation for this volume is not required .\n\nCloses-Bug: #1946618\nChange-Id: I9729b83be7568928cf962174f98d632a6ecf3d15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bdab8f8b93786bbb15e0a96918662bf52da6c219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bdab8f8b93786bbb15e0a96918662bf52da6c219"}]},"branch":"refs/heads/master"},"3e995a1a4052308f62db8376af08cc7e4e3b3679":{"kind":"REWORK","_number":3,"created":"2021-11-30 11:24:54.000000000","uploader":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"ref":"refs/changes/14/817514/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/817514/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/817514/3"}}},"commit":{"parents":[{"commit":"3e048975d0f75b983c69974945f8e5a9792c488a","subject":"Merge \"Correct type annotations for utils.retry()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e048975d0f75b983c69974945f8e5a9792c488a"}]}],"author":{"name":"PrajaktaB","email":"prajakta.ambulkar@ibm.com","date":"2021-11-11 05:38:54.000000000","tz":0},"committer":{"name":"PrajaktaB","email":"prajakta.ambulkar@ibm.com","date":"2021-11-30 11:18:56.000000000","tz":0},"subject":"[SVF] Update correct Error message for add/remove volumes operation in  group-update","message":"[SVF] Update correct Error message for add/remove volumes operation in  group-update\n\n[Spectrum Virtualize Family] Add/Remove volumes (group-update) operation for  group does not show correct error or log message.\nIf a volume is already present in the group and is still passed in the list of volumes(add_volumes_list) to be added to the group ,then\nthe volume is removed from the list directly  without any warning or info message stating the volume is already present in the group.\nUpdating the cinder code to  print a warning message that the given volume already exists hence\nthe add volume operation for this volume is not required .\n\nCloses-Bug: #1946618\nChange-Id: I9729b83be7568928cf962174f98d632a6ecf3d15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e995a1a4052308f62db8376af08cc7e4e3b3679"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e995a1a4052308f62db8376af08cc7e4e3b3679"}]},"branch":"refs/heads/master"},"84def309ae4ed11829b79aa209247888ff3da30a":{"kind":"REWORK","_number":4,"created":"2021-12-10 08:40:29.000000000","uploader":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"ref":"refs/changes/14/817514/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/817514/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/817514/4"}}},"commit":{"parents":[{"commit":"3e048975d0f75b983c69974945f8e5a9792c488a","subject":"Merge \"Correct type annotations for utils.retry()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e048975d0f75b983c69974945f8e5a9792c488a"}]}],"author":{"name":"PrajaktaB","email":"prajakta.ambulkar@ibm.com","date":"2021-11-11 05:38:54.000000000","tz":0},"committer":{"name":"PrajaktaB","email":"prajakta.ambulkar@ibm.com","date":"2021-12-10 08:38:37.000000000","tz":0},"subject":"[SVF] Update correct Error message for add/remove volumes operation in  group-update","message":"[SVF] Update correct Error message for add/remove volumes operation in  group-update\n\n[Spectrum Virtualize Family] Add/Remove volumes (group-update) operation for  group does not show correct error or log message.\nIf a volume is already present in the group and is still passed in the list of volumes(add_volumes_list) to be added to the group ,then\nthe volume is removed from the list directly  without any warning or info message stating the volume is already present in the group.\nUpdating the cinder code to  print a warning message that the given volume already exists hence\nthe add volume operation for this volume is not required .\n\nCloses-Bug: #1946618\nChange-Id: I9729b83be7568928cf962174f98d632a6ecf3d15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/84def309ae4ed11829b79aa209247888ff3da30a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/84def309ae4ed11829b79aa209247888ff3da30a"}]},"branch":"refs/heads/master"},"f0f5370d1897e2de02e934b86022864df1678f11":{"kind":"REWORK","_number":5,"created":"2021-12-29 06:47:08.000000000","uploader":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"ref":"refs/changes/14/817514/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/817514/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/817514/5"}}},"commit":{"parents":[{"commit":"3e048975d0f75b983c69974945f8e5a9792c488a","subject":"Merge \"Correct type annotations for utils.retry()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e048975d0f75b983c69974945f8e5a9792c488a"}]}],"author":{"name":"PrajaktaB","email":"prajakta.ambulkar@ibm.com","date":"2021-11-11 05:38:54.000000000","tz":0},"committer":{"name":"PrajaktaB","email":"prajakta.ambulkar@ibm.com","date":"2021-12-29 06:44:35.000000000","tz":0},"subject":"[SVF] Update correct Error message for add/remove volumes operation in  group-update","message":"[SVF] Update correct Error message for add/remove volumes operation in  group-update\n\n[Spectrum Virtualize Family] Add/Remove volumes (group-update) operation for  group does not show correct error or log message.\nIf a volume is already present in the group and is still passed in the list of volumes(add_volumes_list) to be added to the group ,then\nthe volume is removed from the list directly  without any warning or info message stating the volume is already present in the group.\nUpdating the cinder code to  print a warning message that the given volume already exists hence\nthe add volume operation for this volume is not required .\n\nCloses-Bug: #1946618\nChange-Id: I9729b83be7568928cf962174f98d632a6ecf3d15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f0f5370d1897e2de02e934b86022864df1678f11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f0f5370d1897e2de02e934b86022864df1678f11"}]},"branch":"refs/heads/master"},"944d65e8ee6d54076c2350341b7d5a7925630847":{"kind":"REWORK","_number":6,"created":"2021-12-29 07:05:34.000000000","uploader":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"ref":"refs/changes/14/817514/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/817514/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/817514/6"}}},"commit":{"parents":[{"commit":"3e048975d0f75b983c69974945f8e5a9792c488a","subject":"Merge \"Correct type annotations for utils.retry()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e048975d0f75b983c69974945f8e5a9792c488a"}]}],"author":{"name":"PrajaktaB","email":"prajakta.ambulkar@ibm.com","date":"2021-11-11 05:38:54.000000000","tz":0},"committer":{"name":"PrajaktaB","email":"prajakta.ambulkar@ibm.com","date":"2021-12-29 07:04:49.000000000","tz":0},"subject":"[SVF] Update correct Error message for add/remove volumes operation in  group-update","message":"[SVF] Update correct Error message for add/remove volumes operation in  group-update\n\n[Spectrum Virtualize Family] Add/Remove volumes (group-update) operation for  group does not show correct error or log message.\nIf a volume is already present in the group and is still passed in the list of volumes(add_volumes_list) to be added to the group ,then\nthe volume is removed from the list directly  without any warning or info message stating the volume is already present in the group.\nUpdating the cinder code to  print a warning message that the given volume already exists hence\nthe add volume operation for this volume is not required .\n\nCloses-Bug: #1946618\nChange-Id: I9729b83be7568928cf962174f98d632a6ecf3d15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/944d65e8ee6d54076c2350341b7d5a7925630847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/944d65e8ee6d54076c2350341b7d5a7925630847"}]},"branch":"refs/heads/master"},"d94e0d913d4eff45cf9e932dcbf08c13968439c2":{"kind":"NO_CHANGE","_number":7,"created":"2021-12-30 06:00:26.000000000","uploader":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"ref":"refs/changes/14/817514/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/817514/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/817514/7"}}},"commit":{"parents":[{"commit":"3e048975d0f75b983c69974945f8e5a9792c488a","subject":"Merge \"Correct type annotations for utils.retry()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e048975d0f75b983c69974945f8e5a9792c488a"}]}],"author":{"name":"PrajaktaB","email":"prajakta.ambulkar@ibm.com","date":"2021-11-11 05:38:54.000000000","tz":0},"committer":{"name":"PrajaktaB","email":"prajakta.ambulkar@ibm.com","date":"2021-12-30 05:51:32.000000000","tz":0},"subject":"[SVF] Update correct Error message for add/remove volumes operation in  group-update","message":"[SVF] Update correct Error message for add/remove volumes operation in  group-update\n\n[Spectrum Virtualize Family] Add/Remove volumes (group-update) operation for  group does not show correct error or log message.\nIf a volume is already present in the group and is still passed in the list of volumes(add_volumes_list) to be added to the group ,then\nthe volume is removed from the list directly  without any warning or info message stating the volume is already present in the group.\nUpdating the cinder code to  print a warning message that the given volume already exists hence\nthe add volume operation for this volume is not required .\n\nCloses-Bug: #1946618\nChange-Id: I9729b83be7568928cf962174f98d632a6ecf3d15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d94e0d913d4eff45cf9e932dcbf08c13968439c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d94e0d913d4eff45cf9e932dcbf08c13968439c2"}]},"branch":"refs/heads/master"},"23665c2119a73d8b54e0b52109607977c83c90e7":{"kind":"REWORK","_number":8,"created":"2021-12-30 12:04:54.000000000","uploader":{"_account_id":34106,"name":"Prajakta","email":"prajakta.ambulkar@ibm.com","username":"prajaktab"},"ref":"refs/changes/14/817514/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/817514/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/817514/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/817514/8"}}},"commit":{"parents":[{"commit":"3e048975d0f75b983c69974945f8e5a9792c488a","subject":"Merge \"Correct type annotations for utils.retry()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e048975d0f75b983c69974945f8e5a9792c488a"}]}],"author":{"name":"PrajaktaB","email":"prajakta.ambulkar@ibm.com","date":"2021-11-11 05:38:54.000000000","tz":0},"committer":{"name":"PrajaktaB","email":"prajakta.ambulkar@ibm.com","date":"2021-12-30 10:52:41.000000000","tz":0},"subject":"[SVF] Update correct Error message for add/remove volumes operation in  group-update","message":"[SVF] Update correct Error message for add/remove volumes operation in  group-update\n\n[Spectrum Virtualize Family] Add/Remove volumes (group-update) operation for  group does not show correct error or log message.\nIf a volume is already present in the group and is still passed in the list of volumes(add_volumes_list) to be added to the group ,then\nthe volume is removed from the list directly  without any warning or info message stating the volume is already present in the group.\nUpdating the cinder code to  print a warning message that the given volume already exists hence\nthe add volume operation for this volume is not required .\n\nCloses-Bug: #1946618\nChange-Id: I9729b83be7568928cf962174f98d632a6ecf3d15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23665c2119a73d8b54e0b52109607977c83c90e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23665c2119a73d8b54e0b52109607977c83c90e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
