)]}'
{"id":"openstack%2Fcinder~818884","triplet_id":"openstack%2Fcinder~stable%2Fwallaby~I442a42b0c098775935a799876ad8efbe141829ad","project":"openstack/cinder","branch":"stable/wallaby","topic":"detach-vol-status","attention_set":{},"removed_from_attention_set":{"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-02-02 19:48:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I442a42b0c098775935a799876ad8efbe141829ad","subject":"Delete attachment on remove_export failure","status":"MERGED","created":"2021-11-23 11:52:30.000000000","updated":"2022-02-02 20:05:20.000000000","submitted":"2022-02-02 19:48:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818884-detach-vol-status","meta_rev_id":"f2d54272b4237a247f587b1d398a8ed14408c5d4","_number":818884,"virtual_id_number":818884,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-02 19:48:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-01-28 22:59:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-02 09:24:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-02 09:24:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-23 12:20:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-23 12:48:10.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-23 12:54:54.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-23 13:16:23.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-11-23 14:06:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-23 14:26:18.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-11-23 14:51:48.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-11-24 12:29:53.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 11:36:16.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2022-01-17 16:33:59.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-01-17 18:20:01.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-01-17 22:19:38.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-01-17 22:25:54.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-28 22:59:06.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-02-02 09:24:20.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-02-02 15:07:50.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2022-02-02 17:32:29.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-02 20:05:20.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"}],"messages":[{"id":"3ea1628059c0c3e4cfc1ef345bdec9e565a282da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-11-23 11:52:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8836ad3c561b96da76b8b44cbd4593c20167aac0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 12:20:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31551f8f61ed41b095a56dfce1b50808 : SUCCESS in 22m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea4d422b67c543baa4844f03d14410e1 : SUCCESS in 24m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dcdd0fc66c32cef287589d8789a19ea2077556b2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-23 12:48:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/84/818884/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/43096a2/43096a2.html : FAILURE in 50m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2949cbcea2a846fe9d9c3b0892a63e2873b9962b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-23 12:48:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/84/818884/1/check/iscsi-driver-primera/1756c8e/1756c8e.html : FAILURE in 51m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4f3d3a5ccc47db5886c95da1f94afdb2dfe78eff","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-23 12:54:54.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-84-818884-1/ns5_nfs : FAILURE in 16m 43s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-84-818884-1/ns5_iscsi : FAILURE in 16m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"bc4ac0933dd956e784f86c0b3cc5bdfb292c3429","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-23 13:07:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/84/818884/1/check/iscsi-driver-alletra-6k/66fbc62/66fbc62.html : FAILURE in 51m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f1e1caf1496554d1ad97c142d59440b54454c7b4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-23 13:16:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/818884/1/iscsi/ : SUCCESS in 1h 23m 00s\n- cinder-smb http://cloudbase-ci.com/cinder/818884/1/smb/ : SUCCESS in 1h 05m 31s","accounts_in_message":[],"_revision_number":1},{"id":"03ddfe7a5da64b93d95261f58c3ad0e927e9e732","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 14:06:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8da7d94456c84f6486408e4430cf0489 : SUCCESS in 13m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f93ba90e0152419c936cb1b7223dafb1 : SUCCESS in 9m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b71a98d71b744af9d03df0c9edda838 : SUCCESS in 11m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aff657fd98e046cfb589911b6aaa27e1 : SUCCESS in 14m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a298199bb4a744f9b41081245e9abc89 : SUCCESS in 11m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/17dd6f68b5be43a3952e4df6bfbf9d25 : SUCCESS in 1h 02m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1f7ca7b9abe14b0e998c0290c75ccbf7 : SUCCESS in 1h 07m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37825d7ae865473390d22d599efbe6c8 : SUCCESS in 7m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/390c4afbc30343d5bbc0ccf13040f7b0 : SUCCESS in 14m 14s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b63d88229fba446abd2e26493635d79d : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/452a69eae52e4267a441957cbf64a88c : SUCCESS in 8m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9906341f24244e86a6b51b8db5a4a2f4 : SUCCESS in 6m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/07ee3bcef71b4ea2b9c970f4d903dbf5 : SUCCESS in 58m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eddc416d7e8c415384aaa21c3ac12077 : FAILURE in 7m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e3dbdfe753d4462196832f58da845d0d : SUCCESS in 1h 06m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4bb29771cfbd4e91b0d54402a430519e : SUCCESS in 1h 29m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fbfabda78b944d3999b524600cba62dc : SUCCESS in 1h 42m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b15d1df60bd64c74892baf9d42656572 : SUCCESS in 1h 02m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6c6a372853c24147aa794761bc8fc78a : SUCCESS in 1h 14m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3ab3729f8bad4618b67db84a13cdaa7a : SUCCESS in 1h 33m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cc5b71251604418d853e69f0b63d4263 : SUCCESS in 2h 05m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea0bb2cab97849fb951fddac186db54d : SUCCESS in 47m 46s","accounts_in_message":[],"_revision_number":1},{"id":"872c49f3d8a2e073efd2e5df0898ccfb4d4d8219","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-23 14:26:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/84/818884/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/034731d/ : FAILURE in 2h 32m 15s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/84/818884/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/531768a/ : SUCCESS in 1h 08m 03s","accounts_in_message":[],"_revision_number":1},{"id":"4b7d2f768ccc4d42870a5b659c9c43e2f2bc4148","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-23 14:51:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/84/818884/1/check/dsvm-tempest-cisco-zonemanager-job/8b31fff/ : SUCCESS in 1h 15m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3972c1be4adb1e0c7e5bc99b3b2cb52b46eb996e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-23 17:36:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/84/818884/1/check/iscsi-driver-nimble/44c763d/44c763d.html : FAILURE in 58m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a54e2466722966cad376accde60063a627577a6c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-24 12:29:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/84/818884/1/20211124182958/NEC-ISCSI : SUCCESS in 2h 52m 01s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/84/818884/1/20211124182958/NEC-FC : SUCCESS in 2h 59m 51s","accounts_in_message":[],"_revision_number":1},{"id":"4432a14afe7a0f659bb3f4774ea348bcb61f4e1a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-24 18:59:10.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-84-818884-1/ns5_nfs : SUCCESS in 4h 45m 1s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-84-818884-1/ns5_iscsi : SUCCESS in 4h 35m 34s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"b12af7b3f6c65ae5fa2e2bdd55f1d6536eb2628e","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-11-25 11:36:16.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/84/818884/1/check/dsvm-tempest-TOYOU-ACS5000-driver/50b671c : FAILURE in 37m 48s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/84/818884/1/check/fc-test/b8818c3 : FAILURE in 45m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e14c7f74171828f55a8b47c9b25705bd827314d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-01-17 16:18:48.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b91f4542709144f3f084bf492927da430f2499b8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-01-17 16:21:41.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"9dbb50890077092cc80fe0a854f4199f1c01a444","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-17 16:33:59.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/17_01_2022_16:28:19\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"8036ac5598152ad379b16d45accb6b44fea91dff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 16:45:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/32316895519d4afb91eaed4ef078b695 : SUCCESS in 13m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/64a27d68895146b980bc27ed3b86af3d : SUCCESS in 17m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"43fc4d3dea2a94c653f754bc2d81acbd62dffce4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-17 17:52:12.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/818884/2/iscsi/ : SUCCESS in 1h 01m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/818884/2/smb/ : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":2},{"id":"dca26268b3d4e20fed1a6c82c72c36c15d847b21","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-01-17 18:20:01.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/84/818884/2/check/iscsi-driver-alletra-6k/ce4ebb2/ce4ebb2.html : FAILURE in 21m 47s","accounts_in_message":[],"_revision_number":2},{"id":"fbbfe652a0ba5ce6c7c4085c2f054c0df774aed1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 18:50:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bec1927e0fc4583bb77488096197b91 : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3385c823dfe943cfbee23a6ed0d975a6 : SUCCESS in 11m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8b04d92e9334686ba1908705d31bc03 : SUCCESS in 11m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93842883f64a4ae6b971debee350d076 : SUCCESS in 13m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b8828b45a234683ba208db3410551cd : SUCCESS in 13m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/62e91378b4bb44dcaf58eb7d1b7b25ec : SUCCESS in 59m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f4638cd3eab84f8f9cea1afc9dd90558 : SUCCESS in 1h 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dda0f6e63ded4e33a6ff7e6b62b158c0 : SUCCESS in 54m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/735f66cf6b614b8a832dcf6b9af68600 : SUCCESS in 8m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c0a4f0d4cdf743c5ac0a077e4cd2f29a : SUCCESS in 13m 54s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/085045695d0e42b4a72f9d85fa47e187 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e2c20e17e4994df1aff55ca0dafcb22d : SUCCESS in 7m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c762721a728f442a9515654b40645428 : SUCCESS in 8m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/af82cda36bb74848ac4e3b82df5b2ee2 : SUCCESS in 46m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b180460fe7554fa4b3488a437b34ed93 : FAILURE in 7m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/563ddd0784fa442f975029ce5edca5bf : SUCCESS in 1h 21m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9d1478f913df4edaad66d64e321befc4 : SUCCESS in 1h 19m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b68e6e5ff3aa48b4841331e12ecff48e : SUCCESS in 1h 35m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/615134945cc94f3fbc3054596e1f1a23 : SUCCESS in 1h 13m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4d548ad1732b467aac3068ef62ab3237 : SUCCESS in 1h 24m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/68a837247d684b85b2035f61db3ed2b9 : SUCCESS in 1h 28m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6ebd5b21102e447b843a663ade784396 : SUCCESS in 2h 24m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9132bf332eb4499e805e779b2139f624 : SUCCESS in 1h 00m 40s","accounts_in_message":[],"_revision_number":2},{"id":"2685082727d16c05cacd7d9bdfcec28a466d5ec6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-17 19:08:42.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/17_01_2022_19:03:43\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"7e9f6107f4cb583fb46a7ea1f9fd848d6e993614","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-01-17 22:19:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-84-818884-2/1642445450 : SUCCESS in 46m 49s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"fea1ff61d87947253c37b1f2af595d8f5620ab02","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-01-17 22:25:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/84/818884/2/ : SUCCESS in 1h 39m 27s","accounts_in_message":[],"_revision_number":2},{"id":"f5ebb6d9b2c3eb90a212eba6af319006dde28135","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-01-18 01:13:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/84/818884/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/30c6b26/ : TIMED_OUT in 3h 03m 04s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/84/818884/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/eef5c7b/ : SUCCESS in 1h 12m 48s","accounts_in_message":[],"_revision_number":2},{"id":"e62e611658c1a0165cf1e9d3eff8b7a2e74710d7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-18 09:19:23.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-84-818884-2/ns5_nfs : SUCCESS in 6h 22m 54s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-84-818884-2/ns5_iscsi : SUCCESS in 5h 47m 7s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"4f605c632d10ac66727fa7faa31556a17110c8cd","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-01-19 21:45:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/84/818884/2/20220120004838/NEC-FC : SUCCESS in 2h 48m 46s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/84/818884/2/20220120004838/NEC-ISCSI : SUCCESS in 5h 56m 32s","accounts_in_message":[],"_revision_number":2},{"id":"bdb20b13cf2c5ffb20982b1a260711a5b6481d38","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-01-28 22:59:06.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f66b043d210cadffbd9cb5cd8adc4f97386ea139","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-02-02 09:24:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"de2e733da9926c7bcc95c561ccdf4b646add3121","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 09:27:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"df43deeff5322251d6ee529ca955c19144fab6da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 11:02:19.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/811cff9a7c4a4f5686386068dc93ffc3 : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/211cd0e61ecf4147af5387e23b8cb9e3 : SUCCESS in 11m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b194da2378b4f52a307ae243ef33e64 : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f838b1c7c42948a6817a794166f2389a : SUCCESS in 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/08acaece35af42cd8a97d154ab686d44 : SUCCESS in 58m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c036a0d48f6946e49580c5906f592fcf : SUCCESS in 1h 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d18760c0b3244df683738b6625c186d2 : SUCCESS in 54m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9002005d42ac4ab6b64791950c927cc7 : SUCCESS in 7m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5f4b97928ae046329c3796188e62aea8 : FAILURE in 1h 12m 04s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9d8948673bdd4019a0f69690a7b0ff90 : SUCCESS in 1h 17m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/126a62e86f30429d8f389b482122e1c4 : SUCCESS in 54m 28s","accounts_in_message":[],"_revision_number":2},{"id":"000370599e59384424d28da9153d80cf50299de5","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-02-02 13:28:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8bdde20635077ab2e9f5b2b6eb222467510a4a59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 13:59:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/27a5d76af1764e5db79a6cc143f4f829 : SUCCESS in 17m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c721b5abedef49f3b1a9f92c6b54ae48 : SUCCESS in 26m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"73fedb77f9484479be594c548475b4e036366f4f","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-02-02 15:07:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/5ed3ad4/index.html : FAILURE in 1h 38m 15s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/11ba2dc/index.html : FAILURE in 1h 37m 04s","accounts_in_message":[],"_revision_number":2},{"id":"66ef28bbfdd4943c13050d79cb4de845e123454c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-02 15:19:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/818884/2/iscsi/ : SUCCESS in 1h 12m 18s\n- cinder-smb http://cloudbase-ci.com/cinder/818884/2/smb/ : SUCCESS in 1h 50m 13s","accounts_in_message":[],"_revision_number":2},{"id":"04c0a792fc45694437148713c3bfec84dd306ac0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-02 15:21:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/84/818884/2/check/dsvm-tempest-cisco-zonemanager-job/3050d09/ : SUCCESS in 1h 38m 22s","accounts_in_message":[],"_revision_number":2},{"id":"eb2e0b3c5626f9a1e5886d13cdd0f69c7ee1a1c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 15:58:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1cd778bd40d4be0a4e1fa98364e9331 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ff27fe3049a4e3f8a31a2793fae803a : SUCCESS in 9m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6623cd0dacea4f26b3b78940bc136a33 : SUCCESS in 10m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d07091ae2414c40a37a39a8cb067004 : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbe78becf0054fea9b164da1acacd7fe : SUCCESS in 12m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/141f27c6f57145d39a53fd4efa065af9 : SUCCESS in 57m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fc3eff484de747bc842c23c4f8d4b316 : SUCCESS in 1h 05m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54eb1b4c42de4435b15d348f49efcdc5 : SUCCESS in 51m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f08ae82aabc942ae8c778fc45b527fa5 : SUCCESS in 7m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/df16a04f958e4b218268c63d3dda2c88 : SUCCESS in 13m 18s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/72171084e310404db06efc4ba8798cfe : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/344471038c8d449abe0a266f0e547343 : SUCCESS in 6m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8d266ede97b14afe8a78290c4dbe3647 : SUCCESS in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e3a7086f627348faa5c712ac6ac4ee4c : SUCCESS in 43m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bb506f7a8a0e40f78a07112e7349d3ce : FAILURE in 5m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/05bf546e50854566ac6c5b372f5deb78 : SUCCESS in 1h 35m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9c5caff1c67945bf91c6d39bf81e2496 : SUCCESS in 1h 31m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5790f19478cf4e52860a6b4c164441ef : SUCCESS in 1h 36m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6eebb511414348ec9a5ec5485ac82c96 : SUCCESS in 1h 17m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cacea7e3d5b3486c88d28707c60d9939 : SUCCESS in 1h 23m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6e77cf901356483c934286f8d43ba2a3 : SUCCESS in 1h 24m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/16b91edc806d4d44a1ba70ed82e777fc : SUCCESS in 2h 21m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/357b04c2f32942e8b8698c98879d6b24 : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":2},{"id":"d70f635aad71e75a191e69cc9e82abae81afacdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 16:02:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7d281bee5ffdcddf470d84c9b1ffc67fed000259","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-02 17:32:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/818884/2/upstream-check/cinder-cDOT-iSCSI/a529352/ : FAILURE in 1h 00m 53s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/818884/2/upstream-check/cinder-cDOT-NFS/2dc514c/ : FAILURE in 35m 05s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/818884/2/upstream-check/cinder-cDOT-FCP/1054d2b/ : FAILURE in 1h 33m 30s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"719b7077e48a18602f3383ebbc4c60d18c1a5460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 19:48:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ff81d6340594ea799f2509f2a213e44 : SUCCESS in 8m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b155c9dd17304843b1d6486f717bbfef : SUCCESS in 12m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f459ba673234db886cee4f19e59f642 : SUCCESS in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1d3194eff1d49d98a16d9d8eda4dfeb : SUCCESS in 13m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/c9dab1d2c34b4ee889e2729e6f077119 : SUCCESS in 1h 03m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/71f01eb1887a416781a9645aa130b073 : SUCCESS in 1h 13m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f90e779d203f4606a0d2caf081396a8a : SUCCESS in 43m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f119872d2a44e05a609116831887b4c : SUCCESS in 9m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/22f14bd63bc64226ac817612079c12c6 : SUCCESS in 1h 08m 59s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f450023a7bd44fe1973727fa09509fc6 : SUCCESS in 1h 17m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83e58d15f9b34fb498bec131dab28f68 : SUCCESS in 1h 01m 37s","accounts_in_message":[],"_revision_number":2},{"id":"d1efd13fbdf4ff4a313b4a97ce1293f1ec94281e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 19:48:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ea4f6a277f8bd7b794cac18395c4a11a3d1df07b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 19:52:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f4a10aaa30e4614933d712e130e0c14 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74872515552c4f71b0d9efd5cbf75b9a : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2},{"id":"f2d54272b4237a247f587b1d398a8ed14408c5d4","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-02-02 20:05:20.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/818884/2/upstream-solidfire-check/cinder-solidfire-iSCSI/bec464a/ : FAILURE in 4h 06m 06s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"be7b001290ab56632aaf5d108aeccb367adea55c","revisions":{"51af88353ebefad5002f44c1c7905ada49e9e9c0":{"kind":"REWORK","_number":1,"created":"2021-11-23 11:52:30.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/84/818884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/84/818884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/84/818884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/84/818884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/84/818884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/84/818884/1"}}},"commit":{"parents":[{"commit":"b90caaaa3f83521e5cc9057ee85f1ea25d58c781","subject":"Fix detach notification","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b90caaaa3f83521e5cc9057ee85f1ea25d58c781"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-07-20 15:15:54.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-23 11:25:58.000000000","tz":60},"subject":"Delete attachment on remove_export failure","message":"Delete attachment on remove_export failure\n\nWhen deleting an attachment, if the driver\u0027s remove_export or the\ndetach_volume method call fails in the cinder driver, then the\nattachment status is changed to error_detaching but the REST API call\ndoesn\u0027t fail.\n\nThe end result is:\n- Volume status is \"available\"\n- Volume attach_status is \"detached\"\n- There is a volume_attachment record for the volume\n- The volume may still be exported in the backend\n\nThe volume still being exported in the storage array is not a problem,\nsince the next attach-detach cycle will give it another opportunity for\nit to succeed, and we also do the export on volume deletion.\n\nSo in the end leaving the attachment in error_detaching status doesn\u0027t\nhave any use and creates confusion.\n\nThis patch removes the attachment record when on an attachment delete\nrequest if the error happens on remove_export or detach_volume calls.\n\nThis doesn\u0027t change how the REST API attachment delete operation\nbehaves, the change is that there will not be a leftover attachment\nrecord with the volume in available and detached status.\n\nCloses-Bug: #1935057\nChange-Id: I442a42b0c098775935a799876ad8efbe141829ad\n(cherry picked from commit 3aa00b087884a278e747e839db0707cf39561382)\nConflicts:\n\tcinder/volume/manager.py\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/51af88353ebefad5002f44c1c7905ada49e9e9c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/51af88353ebefad5002f44c1c7905ada49e9e9c0"}]},"branch":"refs/heads/stable/wallaby"},"be7b001290ab56632aaf5d108aeccb367adea55c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-01-17 16:18:48.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/84/818884/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/84/818884/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/84/818884/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/84/818884/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/84/818884/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/84/818884/2"}}},"commit":{"parents":[{"commit":"ed06fc7452699d76c22b288e1cc8e8bc2afa1849","subject":"Fix detach notification","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ed06fc7452699d76c22b288e1cc8e8bc2afa1849"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-07-20 15:15:54.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-01-17 16:16:49.000000000","tz":60},"subject":"Delete attachment on remove_export failure","message":"Delete attachment on remove_export failure\n\nWhen deleting an attachment, if the driver\u0027s remove_export or the\ndetach_volume method call fails in the cinder driver, then the\nattachment status is changed to error_detaching but the REST API call\ndoesn\u0027t fail.\n\nThe end result is:\n- Volume status is \"available\"\n- Volume attach_status is \"detached\"\n- There is a volume_attachment record for the volume\n- The volume may still be exported in the backend\n\nThe volume still being exported in the storage array is not a problem,\nsince the next attach-detach cycle will give it another opportunity for\nit to succeed, and we also do the export on volume deletion.\n\nSo in the end leaving the attachment in error_detaching status doesn\u0027t\nhave any use and creates confusion.\n\nThis patch removes the attachment record when on an attachment delete\nrequest if the error happens on remove_export or detach_volume calls.\n\nThis doesn\u0027t change how the REST API attachment delete operation\nbehaves, the change is that there will not be a leftover attachment\nrecord with the volume in available and detached status.\n\nCloses-Bug: #1935057\nChange-Id: I442a42b0c098775935a799876ad8efbe141829ad\n(cherry picked from commit 3aa00b087884a278e747e839db0707cf39561382)\nConflicts:\n\tcinder/volume/manager.py\n(cherry picked from commit 1328c68a8cc6db0e4bd6b145daa788570908ed36)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be7b001290ab56632aaf5d108aeccb367adea55c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be7b001290ab56632aaf5d108aeccb367adea55c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
