)]}'
{"id":"openstack%2Fcinder~819200","triplet_id":"openstack%2Fcinder~master~I9fe6187ea8c849812a508b48b342226e50fca831","project":"openstack/cinder","branch":"master","topic":"yoga-python-testing","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2021-12-17 18:45:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9fe6187ea8c849812a508b48b342226e50fca831","subject":"Updating python testing as per Yoga testing runtime","status":"MERGED","created":"2021-11-24 23:43:33.000000000","updated":"2021-12-17 18:50:26.000000000","submitted":"2021-12-17 18:45:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819200-yoga-python-testing","meta_rev_id":"25e5a2c966c62215284e1b580d7e377e6134dc45","_number":819200,"virtual_id_number":819200,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-17 18:45:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-12-14 17:19:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-12-17 06:50:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2021-12-17 14:06:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2021-12-17 14:57:43.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"date":"2021-12-16 17:12:59.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2021-12-17 14:57:43.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-25 00:12:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 00:39:39.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-11-25 01:01:46.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 01:26:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-25 01:31:00.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-11-25 01:42:42.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-11-25 02:17:48.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-11-25 02:35:37.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-11-25 02:38:45.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 03:19:17.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-11-25 03:19:56.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 04:07:14.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 04:15:48.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 04:50:18.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-11-25 05:05:53.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-25 07:27:03.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-11-25 13:33:46.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-26 15:21:34.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-11-27 04:17:23.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-27 06:23:58.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-30 12:34:51.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-14 03:57:51.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-14 04:20:44.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2021-12-14 05:48:59.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-14 05:55:51.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2021-12-14 09:02:52.000000000","updated_by":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"reviewer":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"state":"CC"},{"updated":"2021-12-14 12:11:19.000000000","updated_by":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"reviewer":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"state":"CC"},{"updated":"2021-12-14 17:19:24.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-12-16 19:47:01.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-12-16 20:35:35.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-12-17 06:50:33.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-12-17 14:06:25.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2021-12-17 14:40:21.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2021-12-17 14:57:43.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"}],"messages":[{"id":"6cc6bcba4cc7a5d9186fb06d8b1a28c4f1b7a9ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-24 23:43:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b3008fbe0219e40d472189852fb675304c959f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 00:12:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f455fd0687cb4e34895a15889ac891ce : SUCCESS in 18m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0856e63ee688404e9dc1fa50c7d9256b : SUCCESS in 26m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee217add28ff0c13b4cdebe11a49970351ff2a85","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-25 00:39:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819200/1/iscsi/ : SUCCESS in 55m 27s\n- cinder-smb http://cloudbase-ci.com/cinder/819200/1/smb/ : SUCCESS in 55m 37s","accounts_in_message":[],"_revision_number":1},{"id":"4eacba7ca9c2ae8700767ded9d18af07ea600aa7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-25 01:01:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/00/819200/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/885eed6/885eed6.html : SUCCESS in 1h 16m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e695fcf26bb2b6fc7cba7f474087ef376c325c94","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-25 01:03:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/00/819200/1/check/iscsi-driver-primera/17987a5/17987a5.html : SUCCESS in 1h 17m 54s","accounts_in_message":[],"_revision_number":1},{"id":"39c905a6c22373518b11cca2e0c43c94fa41fdb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 01:26:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e3b7416960c348689ea1c2179dac92c4 : SUCCESS in 2m 22s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6aa31d88c4a1470ba9ee26ab22ed4bd6 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc22fdda66a14a9896378430a3568ec3 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/280bbe61544949d29f46c151283aef78 : SUCCESS in 9m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bad6cd85f79e41f2baf36b68ea7f1ce9 : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47a3c3103d654114a78dd308ba6f1ff2 : SUCCESS in 10m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/55cbece43e574a0fa56455b720b2c5f8 : SUCCESS in 1h 00m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87f4e08696b644b0bc031bd3ab659191 : SUCCESS in 1h 06m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/719b067965754e96a4fa8999d3f3d2da : SUCCESS in 6m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e5067eb7a6e347c6a7ec8e91eb3a1d53 : SUCCESS in 12m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1b555787462642d3bd0dc973d3fdefba : SUCCESS in 5m 39s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/11c97d7bee0e49049f9cfd4ad5ecc455 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4a4ba0715c9f44eeb1a60192de4e54c8 : SUCCESS in 5m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7ffe0cb7ff19432b996f209054c0735b : SUCCESS in 7m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b73dbe8ad141420ba49bc655f00e332b : SUCCESS in 35m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7a3975da74dd4e9c8dbe901b1af76641 : SUCCESS in 5m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fd583f4ddde24547ba3f9d835cbfa82f : SUCCESS in 1h 10m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/31587f76685349748592f4b0464d50a5 : SUCCESS in 1h 19m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d477d88d844a4a2a8a24b654862d81e5 : SUCCESS in 1h 31m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ea8caa8b562445fa9a75f8589e514f26 : SUCCESS in 1h 02m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8b9dc903e9b94212b53040dbc79ccf15 : SUCCESS in 1h 11m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/36407ea13cf54ec6a8595b62c1f77dc0 : SUCCESS in 36m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/881a1dc448d34f50a46d507432748808 : SUCCESS in 1h 14m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4512e4aa3fe74950a2e8fa9b68529904 : SUCCESS in 1h 39m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea5f85c126b0419b95af2ebfa822c516 : SUCCESS in 52m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c6ee77083ac4fc7a6b35a5419097bd8 : SUCCESS in 49m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5427cfd9a81c82930fac38742ad5bd942f15e300","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-25 01:31:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/00/819200/1/check/dsvm-tempest-cisco-zonemanager-job/cfd5116/ : SUCCESS in 1h 09m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b8cddfb7c937b13f196c5115a4072e26efd2dfc2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-11-25 01:42:42.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/00/819200/1/check/DellEMC_Unity_FC/430b999/DellEMC_Unity_FC/190 : SUCCESS in 1h 05m 53s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/00/819200/1/check/DellEMC_Unity_iSCSI/70d6459/DellEMC_Unity_iSCSI/192 : FAILURE in 22m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"8ae9dec342197f3cc15f6268873797536fa27b8f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-25 02:17:48.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/00/819200/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2cb9859/ : FAILURE in 2h 29m 19s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/00/819200/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/85ae41f/ : FAILURE in 2h 33m 02s","accounts_in_message":[],"_revision_number":1},{"id":"6086ec080ce2343cfdfababa2976d25069082555","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-11-25 02:35:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/00/819200/1/upstream-solidfire-check/cinder-solidfire-iSCSI/21f7d6d/ : SUCCESS in 1h 07m 37s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"42764a145a8155ff22e39aca829dcf1973045e07","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-25 02:38:45.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/819200/1/check/tempest-dsvm-full-focal-py3/f2a518c/ : SUCCESS in 2h 08m 56s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/819200/1/check/devstack-plugin-ceph-dsvm-focal-py3/8fd9b94/ : SUCCESS in 2h 48m 54s","accounts_in_message":[],"_revision_number":1},{"id":"2c52345830f12fcad8cbdece815852eaf9bae39d","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-11-25 03:19:17.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/00/819200/1/check/DellEMC_SC_FC/c84f769/DellEMC_SC_FC/66 : UNSTABLE in 1h 00m 37s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/00/819200/1/check/DellEMC_SC_iSCSI/3aa16ec/DellEMC_SC_iSCSI/65 : SUCCESS in 1h 52m 17s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"a3cc11aa1f09a2aee407fc94fcd5d1a493202b88","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-11-25 03:19:56.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/00/819200/1/check/DellEMC_PowerFlex/7ddcc9c/DellEMC_PowerFlex/146 : FAILURE in 1h 52m 54s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"bcffc18c88d25d1e76001f3b4f8382628346f985","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-25 04:07:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/00/819200/1/upstream-check/cinder-cDOT-iSCSI/7ad258e/ : SUCCESS in 1h 32m 51s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/00/819200/1/upstream-check/cinder-cDOT-NFS/4f7a20f/ : SUCCESS in 37m 56s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/00/819200/1/upstream-check/cinder-cDOT-FCP/e09b63e/ : FAILURE in 2h 39m 32s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"7188f521fed026cde97543fa13a096aae15ebcb2","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-25 04:15:48.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/00/819200/1/check/DellEMC_XtremIO_FC/d409dc0/DellEMC_XtremIO_FC/231 : SUCCESS in 1h 10m 17s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/00/819200/1/check/DellEMC_XtremIO_iSCSI/3649084/DellEMC_XtremIO_iSCSI/248 : SUCCESS in 1h 43m 48s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"fd5493c8c9844a2dc55fd3f2a2309fb014125ff5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-25 04:42:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/00/819200/1/check/iscsi-driver-nimble/31ac231/31ac231.html : SUCCESS in 1h 49m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3f4f2ba0ed07d687c5df34ff57e477907804a64b","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-11-25 04:50:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/819200-1/HPMSA-iSCSI : SUCCESS in 1:11:49\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/819200-1/HPMSA-FC : SUCCESS in 43:06.76","accounts_in_message":[],"_revision_number":1},{"id":"84bfa2014da36fcb2699257f1afbeb46e10a7500","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-11-25 05:05:53.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/00/819200/1/check/DellEMC_VNX_iSCSI/17fa758/DellEMC_VNX_iSCSI/132 : FAILURE in 3h 03m 27s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/00/819200/1/check/DellEMC_VNX_FC/44b2b83/DellEMC_VNX_FC/133 : FAILURE in 24m 12s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"22656714b7bf4dbf028981061dd6b1f100ff64af","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-11-25 07:27:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/819200-1/Lenovo-iSCSI : SUCCESS in 1:30:32\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/819200-1/Lenovo-FC : SUCCESS in 58:52.02","accounts_in_message":[],"_revision_number":1},{"id":"3b7b393d90b355853787d6a55b192b9c0f184122","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-11-25 13:33:46.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/00/819200/1/check/DellEMC_PowerMAX_FC/9bb895e/DellEMC_PowerMAX_FC/74 : SUCCESS in 3h 35m 38s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/00/819200/1/check/DellEMC_PowerMAX_iSCSI/e48228e/DellEMC_PowerMAX_iSCSI/72 : FAILURE in 3h 21m 39s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"db12c9dc3bfc1bbdde3dcbc2663cc5ad70210044","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-11-26 15:21:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/00/819200/1/check/dsvm-tempest-TOYOU-ACS5000-driver/93e7922 : SUCCESS in 2h 07m 28s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/00/819200/1/check/fc-test/33805c1 : SUCCESS in 2h 05m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a98aca874f473823b85dd14fc98f1a6447202e6b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-26 15:42:55.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f223264b927be1c4523f7b4d0382d300a63fb9ae","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-27 04:17:23.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-00-819200-1/ns5_nfs : SUCCESS in 4h 41m 38s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-00-819200-1/ns5_iscsi : SUCCESS in 4h 48m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"6cac0be8b9a3ab39262046064ab33cbdc39e4601","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-27 06:23:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/819200/1/20211127123002/NEC-ISCSI : SUCCESS in 2h 42m 58s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/819200/1/20211127123002/NEC-FC : SUCCESS in 2h 53m 53s","accounts_in_message":[],"_revision_number":1},{"id":"72355deef6baf2f0bec3766c3996f570ec3fb50d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-11-30 12:34:51.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/819200/819200/1/check-not-vote/ext-cinder-zuul/bee7f88 : SUCCESS in 31m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a79f68678998b0968d102f41405bdb18691326f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-12-14 02:31:27.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4e6b955c27c0cecbdaf3b2c79f5529378f3a872c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 03:14:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/09a32424d68d405aa866ebc5c5cd9792 : SUCCESS in 23m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a96f57e7f5f4779a99a9e1b45878706 : SUCCESS in 30m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e63953462296d469cce446a01f4d6aa9a55395a5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-14 03:17:39.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/819200/2/check/tempest-dsvm-full-focal-py3/d7fbf95/ : RETRY_LIMIT in 11m 06s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/819200/2/check/devstack-plugin-ceph-dsvm-focal-py3/62d8f89/ : RETRY_LIMIT in 10m 45s","accounts_in_message":[],"_revision_number":2},{"id":"532bfff27c9b0b3c675056e4b948af30acfcfa20","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-14 03:36:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819200/2/iscsi/ : SUCCESS in 1h 04m 35s\n- cinder-smb http://cloudbase-ci.com/cinder/819200/2/smb/ : SUCCESS in 55m 01s","accounts_in_message":[],"_revision_number":2},{"id":"84d97ebee9e255afa99b05bb61ce98ba3d1518fa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-14 03:57:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f84065dddfc74b2496a71da405811e59 : SUCCESS in 1h 25m 12s","accounts_in_message":[],"_revision_number":2},{"id":"fac32784fe6a571fb947b086c6e85c7fb6432365","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-14 04:20:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/00/819200/2/check/iscsi-driver-alletra-6k/a7b6f4c/a7b6f4c.html : SUCCESS in 1h 42m 26s","accounts_in_message":[],"_revision_number":2},{"id":"e93000115ccfb79c842bf755e62c2841fad110bd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-14 05:35:47.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/00/819200/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/890de26/ : TIMED_OUT in 3h 03m 08s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/00/819200/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b3f333c/ : SUCCESS in 55m 16s","accounts_in_message":[],"_revision_number":2},{"id":"d2d99577d8212cd82d5b4de185fe8a6f7ef97455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 05:37:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e00a21e2e3a54e6a9fbf93ca174b6eb1 : SUCCESS in 3m 01s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d2fe7da12f3445a28413ec8fca7f829c : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f887e8a0c084905ac0f0a67abefe3b2 : SUCCESS in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e64f6c57d47b44079ac7921e261724af : SUCCESS in 12m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/710090144b7648a9b9a3c848927d57eb : SUCCESS in 13m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/574b97b3b60c487f930b11e2a8c062b5 : SUCCESS in 13m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/87e0e5436c7548aa9637e9fc624a1a7e : SUCCESS in 59m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a02ed80fd8e94c24b8605b60a21f440f : SUCCESS in 1h 09m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef0f5c63287149cb98521a16edf519d6 : SUCCESS in 49m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c533a14e2214b4bb3c63f0b0e76c2dc : SUCCESS in 6m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/51e6b448f8fa4530a38d35562ea608fa : SUCCESS in 14m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/23c04f4bb4d34712966dc24653dc2b75 : SUCCESS in 8m 41s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/22377081bc574cb19c516518399c800c : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a600d4d1d496474f946303a5da0087b4 : SUCCESS in 7m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/065a5d337927484eacb3ac85ca7e23d6 : SUCCESS in 9m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3d93124ae2fa4187985aff4a47f1453f : SUCCESS in 41m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/184a030cb14e40fab4ce7995225680b6 : SUCCESS in 7m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1693d8931d0c445591ebe11d9f199c75 : SUCCESS in 1h 22m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0b37be4af9af44cb9912874aa4f87783 : SUCCESS in 1h 31m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eb01cf2daef64a588c2f3c9497eb5a12 : SUCCESS in 1h 13m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d0bc887025464fda97c8a7a02ea45881 : SUCCESS in 47m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5970c340d5f4472097b0ed4ababbb48b : SUCCESS in 1h 13m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4ca27e7454cc4d69987044a20f263556 : SUCCESS in 40m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1b6df088f4f148af8ef26165d5bf4bc6 : SUCCESS in 1h 16m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8a5c1452b82c44a3b2095ff1657a95e8 : SUCCESS in 2h 20m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b02723713df4b409265e0eb77f802a5 : SUCCESS in 50m 46s","accounts_in_message":[],"_revision_number":2},{"id":"7521d019debc705b606ec09fb8b3ae4457288bbe","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-12-14 05:48:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/00/819200/2/ : SUCCESS in 1h 34m 34s","accounts_in_message":[],"_revision_number":2},{"id":"06e669a529c8083f6d34477daa252a8c3ea7e7d3","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2021-12-14 05:55:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-00-819200-2/1639460259 : SUCCESS in 18m 11s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"5578e1d8fdfa2b09ffe4af754098823ae13cc967","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-14 06:15:52.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/00/819200/2/check/iscsi-driver-nimble/2e17e63/2e17e63.html : SUCCESS in 1h 50m 02s","accounts_in_message":[],"_revision_number":2},{"id":"c66d0eb305bf7ff4f3335cb157b19100faa17f62","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-14 07:24:27.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-00-819200-2/ns5_nfs : SUCCESS in 4h 5m 22s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-00-819200-2/ns5_iscsi : SUCCESS in 4h 8m 5s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"be4e328cdf0a6bbb40edafc5d10b1815286ca3fb","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-12-14 08:03:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/00/819200/2/upstream-solidfire-check/cinder-solidfire-iSCSI/f771de1/ : FAILURE in 23m 43s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"1901b450ab9eef345aa0018e1a44a90eca67c5c5","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-12-14 08:13:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/819200-2/HPMSA-iSCSI : SUCCESS in 1:09:01\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/819200-2/HPMSA-FC : SUCCESS in 43:31.80","accounts_in_message":[],"_revision_number":2},{"id":"4fffb773e316c4b6b9c12e4202e475490af15d35","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-12-14 08:17:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/819200/2/20211214141951/NEC-ISCSI : SUCCESS in 2h 46m 14s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/819200/2/20211214141951/NEC-FC : SUCCESS in 2h 58m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ef9dec769c1a3f920c7aa99f7b99e9a2b0ea255f","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-14 08:58:37.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/00/819200/2/check/dsvm-tempest-TOYOU-ACS5000-driver/6eef3de : FAILURE in 1h 53m 22s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/00/819200/2/check/fc-test/93c09a9 : FAILURE in 2h 07m 23s","accounts_in_message":[],"_revision_number":2},{"id":"43e8c02e2bace28cb306f3ed46507c0b0f5e9cb8","author":{"_account_id":32464,"name":"caiqilong","email":"406454833@qq.com","username":"cccqqqlll"},"date":"2021-12-14 09:02:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"52c193fa6a180134adfbcf87e7a3d2ab95aee8e4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-12-14 10:18:47.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/00/819200/2/upstream-check/cinder-cDOT-iSCSI/93f7dbf/ : FAILURE in 2h 25m 18s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/00/819200/2/upstream-check/cinder-cDOT-NFS/9707b90/ : FAILURE in 1h 42m 50s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/00/819200/2/upstream-check/cinder-cDOT-FCP/530e119/ : FAILURE in 2h 08m 47s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"d4c97a3ea2bfbbc4d42f16f0c9c5a9ef2f0b1ab5","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-14 11:19:52.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/00/819200/2/check/dsvm-tempest-TOYOU-ACS5000-driver/040daa2 : FAILURE in 2h 14m 33s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/00/819200/2/check/fc-test/b01fb4f : FAILURE in 2h 15m 10s","accounts_in_message":[],"_revision_number":2},{"id":"34a34e473c4b98e9a0fd7b5f0f994c3b1e70e4ec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":31854,"name":"IBM GPFS CI","email":"ibmgpci@in.ibm.com","username":"ibmgpci"},"date":"2021-12-14 12:11:19.000000000","message":"Patch Set 2:\n\nBuild Successful. ibm-gpfs-dsvm-tempest-full: https://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/ibm-gpfs-dsvm-tempest-full/169/index.html","accounts_in_message":[],"_revision_number":2},{"id":"b803e2fe64cd559679cae090678a17fc2e1695ea","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-12-14 17:19:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"096488c649c6d12c33f630a100125afcb8ce0a71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 17:40:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e8e56eb71c1371b365f6c89471d10648b5970ae8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 19:08:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5df357f8adec43438920732593017c76 : SUCCESS in 3m 59s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8e2d084af3b241569d8d86322cc8759c : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e612f6d733064a0a8889b8d9361605f8 : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ad1b9f282f145788816f1586ffb7c93 : SUCCESS in 10m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d89d75630f8461a828bea70e3e0707c : SUCCESS in 11m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87850e9f935b4e129b28b6764e3c08b9 : SUCCESS in 12m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/573dec2e08044183b791826a995c160a : FAILURE in 44m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0fde1f4175c0416398d43b9b489d1b5b : SUCCESS in 52m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/872f0836867a4453a3704ba423636ce2 : SUCCESS in 48m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88cc3e6bbd144eb3a7d25fc6d8b8b525 : SUCCESS in 7m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/14ee336583e34e048250cd235d07fe93 : SUCCESS in 54m 00s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bc9258c2f7a54678bb2d30510bbf8808 : SUCCESS in 1h 14m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee828db1b39b46bc95cc04d86a748321 : SUCCESS in 39m 18s","accounts_in_message":[],"_revision_number":2},{"id":"93bc0f30b9ec19aa322303896305f30d3e82fc30","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-12-14 22:02:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e3d5e97a5796d83b36512c0e89a27f7a50d61802","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 22:24:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab04785c0163417aadd8e2c6c3b3ea5f : SUCCESS in 13m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/19eb6b0a947141d79721012dc42d590f : SUCCESS in 16m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bcf605cf9381e40a6a4c9faf3d08ae8c05399d2a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-14 23:01:18.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819200/2/iscsi/ : FAILURE in 58m 08s\n- cinder-smb http://cloudbase-ci.com/cinder/819200/2/smb/ : SUCCESS in 51m 41s","accounts_in_message":[],"_revision_number":2},{"id":"29494e8b0521cf183bec1668237b19413482f9b5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-14 23:22:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5071dbdb92d14cf4a646f597458103c8 : SUCCESS in 1h 18m 32s","accounts_in_message":[],"_revision_number":2},{"id":"9894db5bc95322e87e6427aaf027c8c993b562e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 23:39:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b3f2717903d04336bfe0cda0362f5012 : SUCCESS in 2m 34s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/aa89ef81f70249ed80501c2502c7044b : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a29b936d2404441b87eb9a45bba8534 : SUCCESS in 8m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00dceb56947e43dbad13987e2336a403 : SUCCESS in 10m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/561bedb4e0b14c4b9d00f2076a79cbcd : SUCCESS in 11m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab56d797487c4a4e87f6254dbfa4d6b0 : SUCCESS in 12m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/7afbc73d559e4eb59c8342ef8186d577 : FAILURE in 46m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c7f96a6234e14803bcf2c75d382deaa5 : SUCCESS in 1h 28m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3183706ed6684127b4ff518d7823fd9a : POST_FAILURE in 22m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b20ddbceb30648b1a098f48f79b98282 : SUCCESS in 6m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/498e7329582b48a0a356764d128332f0 : SUCCESS in 10m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1bc56e98fbe74c18a1a387138554b03b : SUCCESS in 6m 08s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1f3c97b39e654cff9cde2c33f6ff1b83 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/791cd6c11c76407a968d550aa2191bd6 : SUCCESS in 6m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2c6bee085d334e808fb2bb51823dda35 : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/80d43b3e07ca46ac8b43408fce781e8f : FAILURE in 20m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/77cefd1c897741d4a654ebcc4cf816ac : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/911b36e0ff47438e9f2cd6edd5b4f39c : POST_FAILURE in 23m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/77cfffc6da09488f8f98cf8d95720d42 : POST_FAILURE in 24m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5811527dc57c447c96411f4cde1835d9 : SUCCESS in 1h 14m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7413006cbaa24492a0174a7bfd6c2ef5 : FAILURE in 55m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e7cc3f4dc9784b1f88a491a52736c66b : SUCCESS in 1h 33m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a459c97e31f04a3c8c50ff1bc4282b3d : SUCCESS in 29m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1c10969d114a4ce5b54372c7ac370439 : SUCCESS in 1h 05m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eb4ab0a8a2c145c3b80b50e34d207a28 : FAILURE in 24m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df2932530efd429e97b74e6160c80f65 : SUCCESS in 1h 10m 10s","accounts_in_message":[],"_revision_number":2},{"id":"a05d2a0e01a967096df1ae1b4d3628e5ef8c0d8e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-15 00:28:19.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/819200/2/check/devstack-plugin-ceph-dsvm-focal-py3/0a9c02e/ : SUCCESS in 2h 21m 04s","accounts_in_message":[],"_revision_number":2},{"id":"c9b04c6926ead17f780ad89a065e8e7804faf173","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-12-15 13:25:43.000000000","message":"Patch Set 2: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4ff2fc3756ee4ceb7ffc92404d4ccfe28adc5496","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 14:01:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/132a7b255aa74ff1ab8037978c7e42e6 : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b674329aea3a4e0f8d9506f345f65386 : SUCCESS in 20m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"92bbceb829f0eee7d83f95c3c0d4987978f7826d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-15 14:45:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819200/2/iscsi/ : SUCCESS in 52m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/819200/2/smb/ : SUCCESS in 49m 16s","accounts_in_message":[],"_revision_number":2},{"id":"3be5371afb3e21a3378d86e2afcf615e4d039986","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-12-15 15:12:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/00/819200/2/check/dsvm-tempest-cisco-zonemanager-job/4da53b9/ : SUCCESS in 1h 03m 44s","accounts_in_message":[],"_revision_number":2},{"id":"4a9d3356fdc1bd78969fa1ed35619e1f06b0f55c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 15:54:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ec3b8fab088d4950a23d6e899735a52a : SUCCESS in 2m 58s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e08672adc79a4efb9a29a910a7db68ae : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90f87bfafaa04d2a879689855b0aaa52 : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/857cd8a7dfa5486cb98c6acf9c0ff71a : FAILURE in 11m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f524d6cb8787423289a915849436db24 : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d22b05e9321b47f9adac805a36368944 : SUCCESS in 14m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/f422b78e23d44043a6d8db03049e312b : SUCCESS in 1h 18m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4af1db23bf9646d78b6c281bc8304eec : SUCCESS in 1h 07m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e255763d750c46eca8cb69011b180c2d : SUCCESS in 46m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b31a326b60674dd68a4cda4759c7de9b : SUCCESS in 7m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fddb28896c1e41f39b19a82180918003 : SUCCESS in 13m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b664dfddc004481a80bd570719992fc8 : SUCCESS in 7m 35s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/227a3c948e174b39a68aad967312e2f8 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a390eb183e394b3aa1973cf755f082c7 : SUCCESS in 7m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e750f099d6584dfb9bd6c960fc9ca2fd : SUCCESS in 9m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/24dbef129913475490b74906ba443864 : SUCCESS in 35m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cb31422c98854ed29292cffcac729161 : SUCCESS in 7m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ff15e1aa1f6f4dff98fbc035bbcff0ed : SUCCESS in 1h 13m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4a1546a0ec1b411e9ae6665e305eb25d : SUCCESS in 1h 28m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ca09ddb5da58401fb95d3584ed3fde2b : SUCCESS in 1h 41m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/120a90bd87dd45dc868feba2a355bff8 : SUCCESS in 1h 25m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/063cf00116ab4a84815a15ae9376b629 : SUCCESS in 1h 14m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/812392eea87f487ebeb02dca6df242da : FAILURE in 43m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2f620573d1bd4237bd91e625551939a6 : SUCCESS in 1h 26m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b129ba1a47814d02aeeb5d12c37eebdf : SUCCESS in 2h 11m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66eed45abf3b41f083b82297e7de13e3 : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":2},{"id":"d4c1ca6fed20fdf581e5085ac6573c7650b31897","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-15 16:47:24.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/819200/2/check/tempest-dsvm-full-focal-py3/b17643c/ : SUCCESS in 2h 28m 36s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"6d60dbed76b9641a3b6847d52943a5bca78d0e0a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-16 02:59:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6531cf4291f14d289c04ee1407bd7401 : SUCCESS in 1h 25m 19s","accounts_in_message":[],"_revision_number":2},{"id":"69ee8f54448fd5a09add3e80190ffa6b310aaa8e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-12-16 17:12:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fe983c1128c4a26ede3e5c5d8a53eab1ae23679c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 17:40:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1c4900cd5eb4b5a801106330c218d66 : SUCCESS in 14m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d36e4d3284594c35ad508914066b6562 : SUCCESS in 25m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7e7a505f7ea31113a8320a01971de26fb325aee8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-16 17:51:46.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"72f66a10fa51102143eac0fd1a0c1e5b55fdee52","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-16 18:20:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819200/2/iscsi/ : SUCCESS in 1h 03m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/819200/2/smb/ : SUCCESS in 1h 07m 05s","accounts_in_message":[],"_revision_number":2},{"id":"4a491eff12b9a2f769ae7f467fc6e6cc50c167ba","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-16 18:57:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8525b7a706e745eaaaa5a916393f187d : SUCCESS in 1h 31m 39s","accounts_in_message":[],"_revision_number":2},{"id":"5142f5c31ddb65eb13fc9132003567fd0a69ed99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 19:39:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c4a99f5351c74d9fa834223459706f5e : SUCCESS in 3m 18s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/339a3f243f90435db67d68711caf74c0 : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dcab6b3ca8f488b8a0aa48e5c19d2bb : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc8e45c7c4784b90a4ac8cc768cc2d72 : SUCCESS in 11m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61dd8a0fdfea4fdc9afe64860bd0ec22 : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a108dbb6013412da42804821dbdd70b : SUCCESS in 13m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/93f43429d4824cbfa4af7f96d35f7243 : SUCCESS in 55m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8922766c5a9544a8a380c02d57142e1b : SUCCESS in 1h 10m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a1abd0a951149b59e3c01313a214014 : SUCCESS in 45m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/025e459c81464895ba2e0c87ea279e27 : SUCCESS in 7m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6ff26cb3ca4345fc85fcc06ad726b4e5 : SUCCESS in 15m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a39bdfa631a34850b9a12f53c0db2791 : SUCCESS in 6m 56s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/55b7a53e69b2434fa9a8211036bd87aa : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ca692d9020384690819f779b647f7920 : SUCCESS in 7m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0860ab37259a4b63b3547ac2d6770584 : SUCCESS in 8m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/82935c7a6b8d4fee922461bdcfbc9294 : SUCCESS in 38m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/834bb39deebf490d921ca0615f4d7d54 : SUCCESS in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ee53756d8e0244ddaa2817d6870e584a : SUCCESS in 1h 26m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a9f98e3704c2485e960027fe51948c16 : SUCCESS in 1h 42m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/451cc94ba5cd423cb00fee4a079d94ed : SUCCESS in 1h 44m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/84cc5b93d4114981a066f2413bbdaf9a : SUCCESS in 1h 07m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/64d49491a3824a05a2b05e2810838c1b : SUCCESS in 1h 24m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/64cc888118f54274a16ff6af5a269c13 : FAILURE in 41m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b854c41bdcf943bb964c4563fdfaf508 : SUCCESS in 1h 34m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2bc13107a60a4a9ab8143b2d32ed21e9 : SUCCESS in 2h 20m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33c04eaa56454b0f92b686893344cfde : SUCCESS in 53m 27s","accounts_in_message":[],"_revision_number":2},{"id":"ab14fd4e39b1f9a5b0565fce2395aa57ce0ec5ab","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-12-16 19:43:28.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"2a4395fb231a9ae94534dedfaf2829fe1a3495d9","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-12-16 19:47:01.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"1903d524b6f14d521d734d735f336c176fa385c2","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-12-16 19:50:36.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"c95dbe1027deaba102ef83c6b1aef65610e6840d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-16 20:32:36.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/00/819200/2/check/DellEMC_Unity_FC/a6accf5/DellEMC_Unity_FC/282 : FAILURE in 52m 45s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/00/819200/2/check/DellEMC_Unity_iSCSI/8a6a5d9/DellEMC_Unity_iSCSI/281 : FAILURE in 12m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"a63cb1931d6982e2ad1169ccaa084acbf07475de","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-12-16 20:35:35.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/00/819200/2/check/DellEMC_PowerStore_iSCSI/22bf4e6/DellEMC_PowerStore_iSCSI/167 : FAILURE in 55m 44s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/00/819200/2/check/DellEMC_PowerStore_FC/5052468/DellEMC_PowerStore_FC/178 : FAILURE in 40m 23s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"5caa1bbe180a1a04b14855d94d8d7f1153c0f9be","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-12-16 20:52:43.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/00/819200/2/check/DellEMC_SC_FC/072f636/DellEMC_SC_FC/123 : FAILURE in 37m 10s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/00/819200/2/check/DellEMC_SC_iSCSI/2db06b2/DellEMC_SC_iSCSI/122 : FAILURE in 1h 12m 51s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"9ae0c1cfba75d0155cfa769cf23adf5b4f2f8f31","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-16 20:53:10.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/00/819200/2/check/DellEMC_XtremIO_FC/ae93b48/DellEMC_XtremIO_FC/329 : FAILURE in 1h 00m 13s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/00/819200/2/check/DellEMC_XtremIO_iSCSI/b4d782c/DellEMC_XtremIO_iSCSI/342 : FAILURE in 1h 13m 19s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"8109374737a58149b0ad9da5852aeacb198fcb8f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-12-16 21:02:10.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/00/819200/2/check/DellEMC_PowerFlex/6ffa5d6/DellEMC_PowerFlex/201 : FAILURE in 1h 22m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"5d29d288d8a956ecda6065d9ab72b116b02ac43d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-12-16 21:07:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/00/819200/2/upstream-check/cinder-cDOT-iSCSI/fd8541f/ : SUCCESS in 39m 20s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/00/819200/2/upstream-check/cinder-cDOT-NFS/dac9be9/ : FAILURE in 1h 27m 20s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/00/819200/2/upstream-check/cinder-cDOT-FCP/95f7018/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"727ed5cdcef580d5a9c654084889de84ab3c26d9","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-12-16 21:36:36.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/00/819200/2/check/DellEMC_VNX_iSCSI/6aa9029/DellEMC_VNX_iSCSI/150 : FAILURE in 1h 56m 47s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/00/819200/2/check/DellEMC_VNX_FC/d16b32c/DellEMC_VNX_FC/157 : FAILURE in 17m 10s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"d9cbbff5c7ae8c0c2cd8bd924b78fb62f03c8c6d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-12-17 05:28:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ccea21d76f981af65a4cf1c666ecf9065ae6e5da","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-12-17 06:50:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1d50bda0a31ca5cefd734b7e33d773fcf44dd2be","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-12-17 07:29:49.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"f5ac8db6e764044df55430edafaa58dd9fdbc8c6","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-12-17 14:06:25.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f49a0192513b22cc4a7390483b5baf71457b7dc4","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2021-12-17 14:40:21.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI http://os-logs.tristero.net/dell-ci/819200-2/PVME-iSCSI : SUCCESS in 58:40.47\n- PVME-FC http://os-logs.tristero.net/dell-ci/819200-2/PVME-FC : SUCCESS in 59:08.77","accounts_in_message":[],"_revision_number":2},{"id":"ea4db84ad196f4093b85a63a2465f321b952eebb","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-12-17 14:57:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3ff80d87145bc36dac8912ab197a282cb8d6ad69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 14:58:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f94e43c6d9c5e23b9fff016aabe5c41b2cec9a8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 18:45:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/864340a24ea84fa39a79c508d75e675f : SUCCESS in 3m 44s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0f9dbdd8ec0245ca9184654d2a0ca68d : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d3fe120017f4d49b1bf6b368f1e1d0c : SUCCESS in 9m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2b493eae9f0485d8734b402900b01fb : SUCCESS in 12m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7aaa52a1b5a4ec798cca9ba976bf7d7 : SUCCESS in 10m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/278f991897c441ad88669218dae41c6c : SUCCESS in 11m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/9edddb9cca4845b88fabf3a2bf022037 : SUCCESS in 56m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/07b5326bc6a741d7aff9e36f2ea6b604 : SUCCESS in 1h 15m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/203ef1fcfaf842eca98abae555395e5d : SUCCESS in 43m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5623cbfdb8cf45bb87f4cfe7c598f116 : SUCCESS in 6m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/88971db3f21345b9b604cdddc22f6c34 : SUCCESS in 1h 18m 53s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/400c51da1137452693ddf21071d98b66 : SUCCESS in 1h 11m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/105df42ee08340cb89c36b1747591b2a : SUCCESS in 50m 01s","accounts_in_message":[],"_revision_number":2},{"id":"261deda286e2034ada2c65b1ac7479faaa8bc7d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 18:45:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"25e5a2c966c62215284e1b580d7e377e6134dc45","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 18:50:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/006f38b7b7a547149b645012a426d67f : SUCCESS in 1m 04s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ecb0fb262774f298c29d03db2b3b17d : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20da383ed2124c68b6793acc25cacb31 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c064de299200771bc01aa689376a239c2775f288","revisions":{"70abaef1515b3df3561558c8102180e31e07e851":{"kind":"REWORK","_number":1,"created":"2021-11-24 23:43:33.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/00/819200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/819200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/819200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/819200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/819200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/819200/1"}}},"commit":{"parents":[{"commit":"97aeffa9f58ec18af74172500ec2bb4c674f0af3","subject":"replace deprecated pyparsing method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97aeffa9f58ec18af74172500ec2bb4c674f0af3"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-24 23:23:06.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-24 23:43:32.000000000","tz":-360},"subject":"Updating python testing as per Yoga testing runtime","message":"Updating python testing as per Yoga testing runtime\n\nYoga testing runtime has been updated with py38 and py39\nas voting and removed the py36 testing. Unit tests update are\nhandled by the job template change in openstack-zuul-job\n\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/818609\n\nthis commit makes other required changes in fucntional jobs and\nupdate the metadata in setup.cfg file.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I9fe6187ea8c849812a508b48b342226e50fca831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70abaef1515b3df3561558c8102180e31e07e851"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70abaef1515b3df3561558c8102180e31e07e851"}]},"branch":"refs/heads/master"},"c064de299200771bc01aa689376a239c2775f288":{"kind":"REWORK","_number":2,"created":"2021-12-14 02:31:27.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/00/819200/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/819200/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/819200/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/819200/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/819200/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/819200/2"}}},"commit":{"parents":[{"commit":"97aeffa9f58ec18af74172500ec2bb4c674f0af3","subject":"replace deprecated pyparsing method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97aeffa9f58ec18af74172500ec2bb4c674f0af3"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-24 23:23:06.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-12-14 02:31:27.000000000","tz":0},"subject":"Updating python testing as per Yoga testing runtime","message":"Updating python testing as per Yoga testing runtime\n\nYoga testing runtime[1] has been updated to add py39\ntesting as voting and as we are testing py3.6 and py3.9\nwe do not need to test py3.7|8 explicitly. Unit tests update are\nhandled by the job template change in openstack-zuul-job\n\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/820286\n\nthis commit makes other required changes in zuul.yaml and\nupdate the classifier in setup.cfg file.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I9fe6187ea8c849812a508b48b342226e50fca831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c064de299200771bc01aa689376a239c2775f288"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c064de299200771bc01aa689376a239c2775f288"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
