)]}'
{"id":"openstack%2Fcinder~819568","triplet_id":"openstack%2Fcinder~master~I87d33f215a064a70cea989d524dc26affeea0c3e","project":"openstack/cinder","branch":"master","topic":"log-warn","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-02-09 02:03:56.000000000","reason":"Brian Rosmaita replied on the change"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-09 03:34:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I87d33f215a064a70cea989d524dc26affeea0c3e","subject":"Use LOG.warning instead of deprecated LOG.warn","status":"MERGED","created":"2021-11-29 05:18:49.000000000","updated":"2022-02-09 10:14:54.000000000","submitted":"2022-02-09 03:34:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819568-log-warn","meta_rev_id":"1f0003ac4ee903eb18d7ee3b2259dbbf0724449c","_number":819568,"virtual_id_number":819568,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-09 03:34:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-02-09 02:03:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-02-09 02:03:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-29 05:43:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-29 07:57:26.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-11-29 08:05:31.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-29 08:07:00.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-11-29 08:13:37.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-11-29 08:22:59.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-29 09:29:51.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-29 11:14:27.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-11-29 11:46:40.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-11-29 11:54:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-29 12:42:42.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-29 12:58:05.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-11-29 14:21:15.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-11-29 14:31:47.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-29 14:53:41.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-11-29 15:23:38.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-30 19:45:48.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-02 11:27:01.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-21 11:11:42.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-01-10 09:24:56.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2022-02-08 23:18:25.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-02-08 23:45:25.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-09 01:09:52.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-02-09 01:38:03.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-02-09 01:42:33.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-09 01:44:41.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-02-09 01:46:24.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-02-09 02:13:37.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-09 04:10:36.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-09 05:56:04.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-02-09 07:22:58.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-02-09 10:14:54.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"}],"messages":[{"id":"b85852aefdcd52f504b58282dbb6d07d3c8911f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 05:18:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bed91b13c6e08ff71afaf20d2cb7e2030a7780ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 05:43:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/222189357ef94f98816b08d53519d77e : SUCCESS in 18m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d24b0c52680c47a197814a421ac93078 : SUCCESS in 22m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0cbc2e253dc0c17ea0adb37181fd8ab1cd403de7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 06:09:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"49cdc0aa724c62010150658d7653759ed2a4d7aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 06:18:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d1ccdf8862715098d11bc62da2b841f9e9eab192","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 06:18:23.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"3997fd8734967152a20bde61f8fc495189f2e2a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 06:46:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8fcf07e1c378d99128c25248ecbd20266e3896a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 06:50:37.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"4ea7cd1baeede91a4a801045e8aa7346418b4f69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 07:11:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/01e15e8c143a45a7ae551838c1f8de7c : SUCCESS in 12m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f879cef2cb743e0830c5475384aef60 : SUCCESS in 16m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5ec5df34209b0ef894bd3173333c9048c0d49a7b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-29 07:57:26.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819568/6/iscsi/ : FAILURE in 1h 06m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/819568/6/smb/ : SUCCESS in 49m 13s","accounts_in_message":[],"_revision_number":6},{"id":"6e61e1d0965921b2e9203ab5b5c0b062a410e3ae","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-29 08:05:31.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/68/819568/6/check/iscsi-driver-nimble/9d15d19/9d15d19.html : FAILURE in 1h 10m 36s","accounts_in_message":[],"_revision_number":6},{"id":"3783d4c83e07b8bc9199a0f3d9d886cd960599a3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-29 08:07:00.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/68/819568/6/check/dsvm-tempest-cisco-zonemanager-job/3f155c0/ : SUCCESS in 1h 09m 54s","accounts_in_message":[],"_revision_number":6},{"id":"39ec59a8e70caa8f5f1388d28b10b1fb6b344acd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-11-29 08:13:37.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/68/819568/6/check/DellEMC_Unity_FC/08009f2/DellEMC_Unity_FC/198 : FAILURE in 1h 02m 28s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/68/819568/6/check/DellEMC_Unity_iSCSI/96f03f6/DellEMC_Unity_iSCSI/200 : SUCCESS in 1h 22m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"e115ada2fce088064534f9fee967249ef1596261","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-29 08:22:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819568/1/20211129141850/NEC-ISCSI : SUCCESS in 2h 33m 28s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819568/1/20211129141850/NEC-FC : SUCCESS in 3h 04m 06s","accounts_in_message":[],"_revision_number":1},{"id":"1e9bd71428e52628f88a2efecee1cd12b3ec97b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 09:21:39.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"93a763fa0f9accb9126677027dfe19782b228b04","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-29 09:25:08.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k finger://openstackcifc-virtual-machine/b48a748423014389a6b462b7cb33ec9e : RETRY_LIMIT in 12s","accounts_in_message":[],"_revision_number":7},{"id":"2bed933f401caf59f6a3b32aaffd056cd08a736a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-29 09:29:51.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819568-1/ns5_nfs : SUCCESS in 4h 10m 19s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819568-1/ns5_iscsi : SUCCESS in 4h 10m 7s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"f96cfe6529c981f28167abadbe0d718767409090","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 09:48:04.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a873b16bc924307b5fafe33893054bb : SUCCESS in 18m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dda4c7a6b7634d2ea843db1bdd663fcf : SUCCESS in 24m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a8152e0802e33d9f6a6d220665a067d19f192673","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-29 10:23:49.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819568/7/iscsi/ : FAILURE in 1h 01m 41s\n- cinder-smb http://cloudbase-ci.com/cinder/819568/7/smb/ : SUCCESS in 50m 51s","accounts_in_message":[],"_revision_number":7},{"id":"2b20444c5fcc68b15cb4f4f102beb32c3d2bd78a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-29 10:32:56.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/68/819568/7/check/iscsi-driver-nimble/bad7845/bad7845.html : FAILURE in 1h 09m 57s","accounts_in_message":[],"_revision_number":7},{"id":"375da923d4485ebb302344208212e77fe3ed99d3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-29 10:35:20.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819568-2/ns5_nfs : SUCCESS in 4h 25m 15s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819568-2/ns5_iscsi : SUCCESS in 4h 22m 27s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"dcc39f5e893b9dc93069066c54b480873e15e2d1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-29 10:35:58.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/68/819568/7/check/dsvm-tempest-cisco-zonemanager-job/80a968e/ : SUCCESS in 1h 09m 12s","accounts_in_message":[],"_revision_number":7},{"id":"72d47d767de091ae7468e2b7965a571404830ae1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-11-29 10:44:32.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/68/819568/7/check/DellEMC_Unity_FC/ee0eb49/DellEMC_Unity_FC/199 : SUCCESS in 1h 08m 49s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/68/819568/7/check/DellEMC_Unity_iSCSI/64751c0/DellEMC_Unity_iSCSI/201 : SUCCESS in 1h 22m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"9c92f050304a8375f465d7e2085b942f38926648","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-29 11:14:15.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819568/6/20211129172303/NEC-FC : SUCCESS in 2h 41m 57s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819568/6/20211129172303/NEC-ISCSI : SUCCESS in 2h 51m 06s","accounts_in_message":[],"_revision_number":6},{"id":"3f8910fd50c9519ca3f68ea3bc2d248065d70736","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-29 11:14:27.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/68/819568/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/099b8df/ : SUCCESS in 1h 51m 45s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/68/819568/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/415cbc8/ : SUCCESS in 57m 07s","accounts_in_message":[],"_revision_number":7},{"id":"12873659786b805a82bf78cb69871e7832f34154","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-29 11:22:40.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819568-3/ns5_nfs : SUCCESS in 4h 21m 29s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819568-3/ns5_iscsi : SUCCESS in 4h 27m 34s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"2457698d4523ecffce5df15611f04be23c9c37a9","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-11-29 11:46:40.000000000","message":"Patch Set 7:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/68/819568/7/check/dsvm-tempest-TOYOU-ACS5000-driver/8c65c53 : SUCCESS in 2h 02m 15s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/68/819568/7/check/fc-test/5c67e26 : SUCCESS in 1h 51m 09s","accounts_in_message":[],"_revision_number":7},{"id":"9066841f160aee5ca4cd648072f2f4d84e39875c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 11:54:44.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3ef7a9b6a621401fa48cee65ec16286c : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7395fae9b834082bc04fc1a4489064e : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4139864ef8d4958930957e006795d5f : SUCCESS in 10m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0828958eb4be4f38b601f6f6c7617712 : SUCCESS in 11m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67b4aa9259d940408f91cefa140c69a7 : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/c4d0c380c3c34a0d883a8012169c3900 : SUCCESS in 51m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/257498d5988b4292b4db2bc0b3870bd9 : SUCCESS in 1h 07m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dae82ad0181e4288b54b09c8ebb0db4c : SUCCESS in 7m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3c4c17cd16e24b438852242e71598546 : SUCCESS in 12m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6d03d254889045328aad5606d942ee7c : SUCCESS in 5m 06s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dccd27c4019f4443a94b811489a3575a : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/93ffde97b13c4dc8af6d14a9311c7767 : SUCCESS in 8m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b6038a8951b245bb80cccc0b5433cfee : SUCCESS in 6m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e819000bafaa4e7fa5d4cf283fcb3710 : SUCCESS in 28m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/61dfd0264ba649d3b2f367c2fb4b4a9b : SUCCESS in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8b89305ff57d43dd950317b3f0895889 : SUCCESS in 1h 51m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/39fe856a00544c78b09529580147e314 : SUCCESS in 2h 14m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1d686d397e0342b797c6fdffc16c71c7 : SUCCESS in 1h 29m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/aa9ebe808d4245f0b19f150f93ee0567 : SUCCESS in 1h 04m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e84165da1dbe42559cd9439b9d73d02c : SUCCESS in 1h 10m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f94e51c3ef614af39c35f6dd2c2b29ad : SUCCESS in 38m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a9a7947d724a46ca8b6f3647581eeee1 : SUCCESS in 1h 11m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/edbbda180bb146b7b0a53a7f57ca7d99 : SUCCESS in 2h 25m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06b6e3cbde4643b38e0c3f9ce7eb101a : SUCCESS in 45m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8664298b6aa4d81bda813f50ff823d9 : SUCCESS in 47m 18s","accounts_in_message":[],"_revision_number":7},{"id":"9058382e620f158be4fda652a87f149e8f227f6b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-29 12:10:25.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819568-4/ns5_nfs : SUCCESS in 4h 27m 51s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819568-4/ns5_iscsi : SUCCESS in 4h 30m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"c21253fd06615169995f727e6608b3ea78dfd6e4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-29 12:42:42.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/68/819568/7/check/tempest-dsvm-full-focal-py3/7d9eb93/ : SUCCESS in 2h 31m 39s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/68/819568/7/check/devstack-plugin-ceph-dsvm-focal-py3/c6240b6/ : SUCCESS in 2h 43m 35s","accounts_in_message":[],"_revision_number":7},{"id":"cd603d3bd781abbf8bb47f14d0dd3fa1f3afa75f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-29 12:54:44.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819568-5/ns5_nfs : SUCCESS in 4h 29m 36s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819568-5/ns5_iscsi : SUCCESS in 4h 27m 52s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"9fa5b0799cb8aeda38580c59f4f87ab4c5ea4d40","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-11-29 12:58:05.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/68/819568/7/upstream-solidfire-check/cinder-solidfire-iSCSI/47ec3c1/ : SUCCESS in 1h 02m 11s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":7},{"id":"fc53417f244d3f3b4ee3af1303620251aed27159","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-11-29 13:20:19.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/68/819568/7/check/DellEMC_Unity_FC/f427a18/DellEMC_Unity_FC/200 : SUCCESS in 1h 08m 42s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/68/819568/7/check/DellEMC_Unity_iSCSI/b712df2/DellEMC_Unity_iSCSI/202 : SUCCESS in 1h 25m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"1f86bc0e0ba3b4c315a562db9159ab6f10cceb17","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-29 13:59:38.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819568-6/ns5_nfs : SUCCESS in 4h 21m 10s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819568-6/ns5_iscsi : SUCCESS in 4h 21m 36s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"18c8e1b19d435cd6196481dbc372a836c3d76113","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-11-29 14:21:15.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/819568-7/HPMSA-iSCSI : SUCCESS in 1:14:00\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/819568-7/HPMSA-FC : SUCCESS in 43:53.04","accounts_in_message":[],"_revision_number":7},{"id":"7ba219b11419afacd6b70ef4ddfff868fc471189","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-11-29 14:31:47.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/68/819568/7/check/DellEMC_VNX_iSCSI/7c1b3bf/DellEMC_VNX_iSCSI/137 : SUCCESS in 2h 36m 40s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/68/819568/7/check/DellEMC_VNX_FC/6edeed7/DellEMC_VNX_FC/138 : FAILURE in 13m 13s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"4d5a7d3effd05668f23bac480450db1af995857c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-29 14:51:34.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819568-7/ns5_nfs : SUCCESS in 4h 13m 53s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819568-7/ns5_iscsi : SUCCESS in 4h 17m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"f4a9c9d8370a988d9fb68dccbff9ade72d1f2516","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-11-29 14:53:41.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/819568-7/Lenovo-iSCSI : SUCCESS in 1:29:14\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/819568-7/Lenovo-FC : SUCCESS in 1:00:08","accounts_in_message":[],"_revision_number":7},{"id":"605eea562cebe5d83c178d5102cb14c0c18a9667","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-11-29 15:23:38.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/68/819568/7/check/DellEMC_PowerMAX_FC/34fa101/DellEMC_PowerMAX_FC/80 : SUCCESS in 3h 28m 37s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/68/819568/7/check/DellEMC_PowerMAX_iSCSI/17ee21f/DellEMC_PowerMAX_iSCSI/78 : SUCCESS in 3h 00m 58s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":7},{"id":"e1781b3dac15f05a56c7d9e5fd3b344ef794ed9b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-29 17:13:11.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819568/7/20211129201418/NEC-ISCSI : SUCCESS in 2h 41m 58s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819568/7/20211129201418/NEC-FC : SUCCESS in 5h 58m 50s","accounts_in_message":[],"_revision_number":7},{"id":"b9c237a6d05086eaa2dcbcafccba009788b3861f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-30 19:45:48.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/68/819568/7/upstream-check/cinder-cDOT-iSCSI/f67e20d/ : SUCCESS in 1h 39m 56s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/68/819568/7/upstream-check/cinder-cDOT-NFS/cad89b3/ : SUCCESS in 1h 41m 42s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/68/819568/7/upstream-check/cinder-cDOT-FCP/a5a2255/ : FAILURE in 2h 29m 00s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"d9c2ff96c69ce4428ca97d5ab691baab501d22ab","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-12-02 11:27:01.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/819568/819568/7/check-not-vote/ext-cinder-zuul/c3f0227 : SUCCESS in 32m 24s","accounts_in_message":[],"_revision_number":7},{"id":"92d70f8789a025dad5faba60084bd214c7a4a57e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-21 11:11:42.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"454815955cfa5e08c8e5622daf06b562ab9f48dd","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-01-10 09:24:56.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"44e28ef05643b843f9bd2ad34301603811e98052","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-08 23:18:25.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fb3416c3237d39da5b82b972ace921951fd6c71c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-08 23:29:13.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3fbfb59375caed7ecc4bc7703572e9d33cf82dbe","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-08 23:32:42.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1e784d0d924cd2f2602a7337970ea675b10f2d1c","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-08 23:45:25.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/68/819568/8/check/DellEMC_XtremIO_FC/55f7156/DellEMC_XtremIO_FC/684 : FAILURE in 15m 00s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/68/819568/8/check/DellEMC_XtremIO_iSCSI/1348995/DellEMC_XtremIO_iSCSI/694 : FAILURE in 15m 07s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"35f2d5fc2b35d90d524ee49d7a6c171a0e1afe3e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-09 00:24:50.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819568/8/iscsi/ : FAILURE in 54m 28s\n- cinder-smb http://cloudbase-ci.com/cinder/819568/8/smb/ : SUCCESS in 55m 05s","accounts_in_message":[],"_revision_number":8},{"id":"d958c7f9072f9d335f8445e1b12bedc8696c0a68","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-02-09 00:52:10.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/68/819568/8/check/iscsi-driver-primera/27592d1/27592d1.html : FAILURE in 1h 19m 18s","accounts_in_message":[],"_revision_number":8},{"id":"fd72782a01f2d3fecca9d2cace0a09440d9fbe7d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-02-09 00:55:05.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/68/819568/8/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/c06df02/c06df02.html : SUCCESS in 1h 21m 46s","accounts_in_message":[],"_revision_number":8},{"id":"56c4ace4b421e07e5f10d93b18cd74cb3d7d72c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 00:58:26.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0f22a9c0f5b49dea037e5077e7ac342 : FAILURE in 25m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/22b886871673479eb7e7adc09ec53ca2 : SUCCESS in 20m 23s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6eb3621e9dc03168e2b08811d5b23b5b77ec96b2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-09 01:01:41.000000000","message":"Patch Set 8:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":8},{"id":"d9f922070cc1d75421e83e48aab1ba1d936b846a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-09 01:09:52.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/68/819568/8/check/iscsi-driver-alletra-6k/4c94277/4c94277.html : SUCCESS in 1h 38m 34s","accounts_in_message":[],"_revision_number":8},{"id":"dd8176e1b6ea13fd0663bff1cffe53b82ba557fd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-09 01:22:00.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/68/819568/8/check/iscsi-driver-nimble/963ce54/963ce54.html : SUCCESS in 1h 51m 25s","accounts_in_message":[],"_revision_number":8},{"id":"7df79daba043b4f03c71f318b2eb9aaf26a53447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 01:29:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/33ab5bb0a06f48828b15cc1895ded2ec : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebfca10dc3cd4a609c9ae53a079d7f02 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b61ff57c76cc4416b17df1904ba2d1d5 : SUCCESS in 9m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d8747e4b1704506a9be276df889cd57 : SUCCESS in 11m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8286b289d361411bac38a5ad029c4123 : SUCCESS in 11m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/3a16cfd726264192abdfe2b1d01b9f38 : SUCCESS in 52m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea1d7d4da015455da8279d37f82bb258 : SUCCESS in 1h 25m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ce09d3d6d4c49d19bc08be565cd1b86 : SUCCESS in 47m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e826d5654d4043139fcbe36358d475cb : SUCCESS in 6m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/76c1585b8a0e434bb429e704c671d29e : SUCCESS in 13m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3002a1db8ed54c45b5a2abc67121e568 : SUCCESS in 6m 00s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/312fc00759ae49809fb4b010238a0222 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cd8cbd28d2514b34b642e74821439209 : SUCCESS in 6m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3fd27da4e38e4469b0290b2322cf43ed : SUCCESS in 7m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b9149cfb15d3420d972a80edfa76c126 : SUCCESS in 35m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/205aed766b1a4b619e14fa2a250640fe : FAILURE in 5m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1ab3bf8b7cc8474fa5110ee3b6ffbbd5 : SUCCESS in 1h 11m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a6e4f7ab16a0465d83fb160cdddf0900 : FAILURE in 21m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7c8b3a58c2f141f8a04eb456799d1ac8 : SUCCESS in 1h 42m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d2635c91df46455e858dbafaa88056cc : SUCCESS in 1h 01m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/31cb1c91dc224fdf91f001869ce0961d : SUCCESS in 1h 01m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bdd7b2e4e5714b38a912f3a64b237067 : SUCCESS in 29m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bc0dd48f57dc435ea91a929f27836fd9 : SUCCESS in 1h 06m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e7b3ec747a2f4956bb2683a777f23146 : SUCCESS in 1h 46m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ad78f72c4bf46e8bf3b9ddbb0c311ba : SUCCESS in 44m 27s","accounts_in_message":[],"_revision_number":8},{"id":"0c6e21e343bccc1af1d62742309fba0378b2b956","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-09 01:31:39.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/819568/8/upstream-check/cinder-cDOT-iSCSI/98ff61a/ : NOT_REGISTERED\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/819568/8/upstream-check/cinder-cDOT-NFS/07c80c6/ : NOT_REGISTERED\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/819568/8/upstream-check/cinder-cDOT-FCP/a5a0800/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"1352ff617bfcffac9342a057456040c9ae4ff5b3","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-02-09 01:31:40.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/819568/8/upstream-solidfire-check/cinder-solidfire-iSCSI/899c45e/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"7117d0ec62ca93358d465837a719f4e2090c830b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-02-09 01:38:03.000000000","message":"Patch Set 8:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/09_02_2022_01_30_07\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"62fa6a0f7f36c061fb9b17ba27816d05ba7ea376","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-09 01:42:33.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/68/819568/8/check/DellEMC_PowerFlex/63b4c3b/DellEMC_PowerFlex/384 : FAILURE in 12m 07s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"971083453df6d7fe085de71cc573c38853163314","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-09 01:43:25.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/68/819568/8/check/devstack-plugin-ceph-dsvm-focal-py3/187450b/ : SUCCESS in 2h 03m 04s","accounts_in_message":[],"_revision_number":8},{"id":"0d19cfa13cfbac3b1010c3519082c9fcf3f0c420","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-02-09 01:44:41.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/68/819568/8/check/DellEMC_PowerStore_iSCSI/8527441/DellEMC_PowerStore_iSCSI/334 : FAILURE in 13m 43s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/68/819568/8/check/DellEMC_PowerStore_FC/8958935/DellEMC_PowerStore_FC/351 : FAILURE in 14m 21s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"c0a7aa46bb74e9e7f9f290b69a6b913ae3cd9f69","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-09 01:45:51.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/68/819568/8/check/DellEMC_XtremIO_FC/743a4b3/DellEMC_XtremIO_FC/689 : FAILURE in 14m 38s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/68/819568/8/check/DellEMC_XtremIO_iSCSI/bdc0ee6/DellEMC_XtremIO_iSCSI/699 : FAILURE in 15m 25s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"8343cf69d538078cd69491ae8bf54c087e5ce04c","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-09 01:46:24.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-68-819568-8/1644370199 : SUCCESS in 16m 22s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":8},{"id":"897004207e1eac82b5eee52a11cc0e311b4707a8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-09 02:03:56.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ccb4120c85f28379b51ebe5017a6cb793a410b7d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-02-09 02:07:53.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/68/819568/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/0489f6e/ : FAILURE in 2h 29m 35s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/68/819568/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/57cfdd2/ : FAILURE in 2h 36m 53s","accounts_in_message":[],"_revision_number":8},{"id":"4066e9be9cc0590fd86297956ee2ed4facb704b1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-09 02:07:58.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/68/819568/8/check/dsvm-tempest-cisco-zonemanager-job/53185f7/ : SUCCESS in 1h 05m 20s","accounts_in_message":[],"_revision_number":8},{"id":"f4b9a5ab6d664da789d53e9839f34a3e24153bf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 02:12:46.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"7d0101b3a8e36dd14000c08ee6242170108092ff","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-02-09 02:13:37.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/68/819568/8/check/dsvm-tempest-zadara-driver/7c95bc3 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 20m 23s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a80fd9330bba3a5a149b218bb1c78bb04538cda7","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-02-09 02:49:22.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/68/819568/8/check/DellEMC_PowerMAX_FC/154ae1b/DellEMC_PowerMAX_FC/296 : FAILURE in 50m 19s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/68/819568/8/check/DellEMC_PowerMAX_iSCSI/6a2f138/DellEMC_PowerMAX_iSCSI/309 : FAILURE in 27m 41s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"ccaddf5ec71ecb6cdc7a95d00f69ecf253150ad6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 03:34:05.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0bd4292661a04a0cac0d7dc5bad75c1c : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23bf61b1098148e5bc05e9e5bb65eefb : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98817b0b14a74133aad8ad2219d016e8 : SUCCESS in 10m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a349f2f7b1924b94b30de4abf33accb4 : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50790806c96749ceb2c1a09316281c7a : SUCCESS in 13m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/f7afae2eb6d34cbd930d42070bb69264 : SUCCESS in 1h 07m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/47aabdf1d8fd414c92b8e6b13ee45c51 : SUCCESS in 1h 06m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d4418fb90ec4d148ec8c7bd59622bc6 : SUCCESS in 51m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e66d44483c04659bf651eb5edce2d67 : SUCCESS in 6m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/720efef58ba4405b8d5912f265a9a562 : SUCCESS in 1h 07m 09s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c7b509cecf7d487fa2f305f979734e71 : SUCCESS in 1h 19m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9a8931d62764997b5b2fa4577581828 : SUCCESS in 56m 34s","accounts_in_message":[],"_revision_number":8},{"id":"3f8e8e6b8fdc01128ba23c8951265f215510da7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 03:34:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"92ca99c7cb3548ce006e0963d4263dc3400e606d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 03:36:10.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/cbe560679eeb4b818bbe9d3964e11750 : SUCCESS in 1m 07s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba721e77f032485c9f17d38e8364aa15 : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c55e190051d4464b4d934fb9c51791b : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":8},{"id":"be626b629d8b15441d4e296bad30c202f3c2a5e4","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-09 04:10:36.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/68/819568/8/ : SUCCESS in 1h 49m 19s","accounts_in_message":[],"_revision_number":8},{"id":"25c03a514a43b4b138870311d0ff4f28b26318b7","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-02-09 05:56:04.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/819568-8/Seagate-iSCSI/ : SUCCESS in 1:03:14\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/819568-8/Seagate-FC/ : SUCCESS in 1:00:42","accounts_in_message":[],"_revision_number":8},{"id":"47b0611025067c22cef1076ba62347894db84dfd","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-09 06:31:59.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/819568-8/HPMSA-iSCSI/ : SUCCESS in 49:58.95\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/819568-8/HPMSA-FC/ : SUCCESS in 42:44.71","accounts_in_message":[],"_revision_number":8},{"id":"0c233d4d88ee318347abe15572278e8018d4094d","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-02-09 07:22:58.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/68/819568/8/check/DellEMC_SC_FC/7756495/DellEMC_SC_FC/318 : FAILURE in 1h 08m 53s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/68/819568/8/check/DellEMC_SC_iSCSI/1457c7b/DellEMC_SC_iSCSI/314 : SUCCESS in 1h 56m 32s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":8},{"id":"6c188b45d0161d816b03168e177d3731cf49661f","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-09 07:52:16.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/819568-8/Lenovo-iSCSI/ : SUCCESS in 1:00:25\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/819568-8/Lenovo-FC/ : SUCCESS in 57:10.57","accounts_in_message":[],"_revision_number":8},{"id":"a2cb84b7f26a4f2533422d4be0f03d38cdcb807d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-09 07:56:37.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819568-8/ns5_nfs : SUCCESS in 4h 31m 47s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819568-8/ns5_iscsi : SUCCESS in 4h 36m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"1f0003ac4ee903eb18d7ee3b2259dbbf0724449c","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-09 10:14:54.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/819568-8/PVME-iSCSI/ : SUCCESS in 1:00:21\n- PVME-FC https://os-logs.seagate.com/dell-ci/819568-8/PVME-FC/ : SUCCESS in 59:44.65","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"a03e10ceb094c222698df0bc90061b70384f5b27","revisions":{"adccca31bbb1766a4b393cd4c251e7c5fa0daf43":{"kind":"REWORK","_number":1,"created":"2021-11-29 05:18:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/819568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819568/1"}}},"commit":{"parents":[{"commit":"3e048975d0f75b983c69974945f8e5a9792c488a","subject":"Merge \"Correct type annotations for utils.retry()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e048975d0f75b983c69974945f8e5a9792c488a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 05:18:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 05:18:00.000000000","tz":540},"subject":"Do not use deprecated LOG.warn","message":"Do not use deprecated LOG.warn\n\n... and use LOG.warning instead.\n\nChange-Id: I87d33f215a064a70cea989d524dc26affeea0c3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/adccca31bbb1766a4b393cd4c251e7c5fa0daf43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/adccca31bbb1766a4b393cd4c251e7c5fa0daf43"}]},"branch":"refs/heads/master"},"d91a06aa81d9b0d7d8bdba7a2bdce980ca8615a2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-11-29 06:09:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/819568/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819568/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819568/2"}}},"commit":{"parents":[{"commit":"97aeffa9f58ec18af74172500ec2bb4c674f0af3","subject":"replace deprecated pyparsing method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97aeffa9f58ec18af74172500ec2bb4c674f0af3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 05:18:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 06:09:27.000000000","tz":0},"subject":"Do not use deprecated LOG.warn","message":"Do not use deprecated LOG.warn\n\n... and use LOG.warning instead.\n\nChange-Id: I87d33f215a064a70cea989d524dc26affeea0c3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d91a06aa81d9b0d7d8bdba7a2bdce980ca8615a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d91a06aa81d9b0d7d8bdba7a2bdce980ca8615a2"}]},"branch":"refs/heads/master","description":"Rebase"},"73453e76bc84daa295b34d4da192764b04ac0a07":{"kind":"REWORK","_number":3,"created":"2021-11-29 06:18:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/819568/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819568/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819568/3"}}},"commit":{"parents":[{"commit":"3e048975d0f75b983c69974945f8e5a9792c488a","subject":"Merge \"Correct type annotations for utils.retry()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e048975d0f75b983c69974945f8e5a9792c488a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 05:18:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 06:13:43.000000000","tz":540},"subject":"Do not use deprecated LOG.warn","message":"Do not use deprecated LOG.warn\n\n... and use LOG.warning instead.\n\nChange-Id: I87d33f215a064a70cea989d524dc26affeea0c3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/73453e76bc84daa295b34d4da192764b04ac0a07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/73453e76bc84daa295b34d4da192764b04ac0a07"}]},"branch":"refs/heads/master"},"ee21d2750bae2803883d0d3894a2c3f1766fb6d7":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-11-29 06:18:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/819568/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819568/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819568/4"}}},"commit":{"parents":[{"commit":"97aeffa9f58ec18af74172500ec2bb4c674f0af3","subject":"replace deprecated pyparsing method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97aeffa9f58ec18af74172500ec2bb4c674f0af3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 05:18:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 06:18:23.000000000","tz":0},"subject":"Do not use deprecated LOG.warn","message":"Do not use deprecated LOG.warn\n\n... and use LOG.warning instead.\n\nChange-Id: I87d33f215a064a70cea989d524dc26affeea0c3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee21d2750bae2803883d0d3894a2c3f1766fb6d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee21d2750bae2803883d0d3894a2c3f1766fb6d7"}]},"branch":"refs/heads/master","description":"Rebase"},"8ca607464db3098d33939f43611c326e5d0a11d6":{"kind":"REWORK","_number":5,"created":"2021-11-29 06:46:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/819568/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819568/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819568/5"}}},"commit":{"parents":[{"commit":"3e048975d0f75b983c69974945f8e5a9792c488a","subject":"Merge \"Correct type annotations for utils.retry()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e048975d0f75b983c69974945f8e5a9792c488a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 05:18:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 06:46:09.000000000","tz":540},"subject":"Do not use deprecated LOG.warn","message":"Do not use deprecated LOG.warn\n\n... and use LOG.warning instead.\n\nChange-Id: I87d33f215a064a70cea989d524dc26affeea0c3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ca607464db3098d33939f43611c326e5d0a11d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ca607464db3098d33939f43611c326e5d0a11d6"}]},"branch":"refs/heads/master"},"47ad155b099ef29c6dd921438a50fb930e8d77ea":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2021-11-29 06:50:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/819568/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819568/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819568/6"}}},"commit":{"parents":[{"commit":"3e048975d0f75b983c69974945f8e5a9792c488a","subject":"Merge \"Correct type annotations for utils.retry()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e048975d0f75b983c69974945f8e5a9792c488a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 05:18:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 06:50:24.000000000","tz":540},"subject":"Do not use deprecated LOG.warn","message":"Do not use deprecated LOG.warn\n\n... but use LOG.warning instead.\n\nChange-Id: I87d33f215a064a70cea989d524dc26affeea0c3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47ad155b099ef29c6dd921438a50fb930e8d77ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47ad155b099ef29c6dd921438a50fb930e8d77ea"}]},"branch":"refs/heads/master"},"96eea87bc5971acee004e7effff794265a7458a1":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":7,"created":"2021-11-29 09:21:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/819568/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819568/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819568/7"}}},"commit":{"parents":[{"commit":"97aeffa9f58ec18af74172500ec2bb4c674f0af3","subject":"replace deprecated pyparsing method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/97aeffa9f58ec18af74172500ec2bb4c674f0af3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 05:18:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 09:21:28.000000000","tz":540},"subject":"Use LOG.warning instead of deprecated LOG.warn","message":"Use LOG.warning instead of deprecated LOG.warn\n\nThe LOG.warn method is deprecated[1] and the LOG.warning method should\nbe used instead.\n\n[1] https://docs.python.org/3/library/logging.html#logging.warning\n\nChange-Id: I87d33f215a064a70cea989d524dc26affeea0c3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96eea87bc5971acee004e7effff794265a7458a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96eea87bc5971acee004e7effff794265a7458a1"}]},"branch":"refs/heads/master"},"a03e10ceb094c222698df0bc90061b70384f5b27":{"kind":"REWORK","_number":8,"created":"2022-02-08 23:29:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/819568/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819568/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819568/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819568/8"}}},"commit":{"parents":[{"commit":"83194a5f4fcc19b08fc9d0d08210c3eecb938dd4","subject":"Merge \"docs: Update \"Getting your patch merged\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/83194a5f4fcc19b08fc9d0d08210c3eecb938dd4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 05:18:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 23:29:02.000000000","tz":540},"subject":"Use LOG.warning instead of deprecated LOG.warn","message":"Use LOG.warning instead of deprecated LOG.warn\n\nThe LOG.warn method is deprecated[1] and the LOG.warning method should\nbe used instead.\n\n[1] https://docs.python.org/3/library/logging.html#logging.warning\n\nChange-Id: I87d33f215a064a70cea989d524dc26affeea0c3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a03e10ceb094c222698df0bc90061b70384f5b27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a03e10ceb094c222698df0bc90061b70384f5b27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
