)]}'
{"id":"openstack%2Fcinder~819668","triplet_id":"openstack%2Fcinder~master~I2a83e9b23f40a8ef4a734b456e4b7afb1ad65f94","project":"openstack/cinder","branch":"master","topic":"new-quota-prework","attention_set":{},"removed_from_attention_set":{"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2024-02-06 01:37:20.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-01-18 13:50:29.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I2a83e9b23f40a8ef4a734b456e4b7afb1ad65f94","subject":"Quota: Add backup related default limits","status":"MERGED","created":"2021-11-29 18:19:05.000000000","updated":"2024-02-06 06:00:08.000000000","submitted":"2024-02-06 01:37:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"819668-new-quota-prework","meta_rev_id":"6fdca83427a9e54def7afdf70f9f164a1c02972b","_number":819668,"virtual_id_number":819668,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-06 01:37:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-02-05 23:34:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-01-24 08:15:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-18 03:27:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2024-01-18 13:50:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-02-05 23:34:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-29 19:00:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-29 20:25:53.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-29 20:35:50.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-11-29 20:46:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-29 21:49:24.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-29 21:56:12.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-29 23:57:19.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-11-30 01:42:30.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-11-30 01:44:38.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-11-30 01:56:01.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-30 02:56:45.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-30 05:13:22.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-30 07:14:33.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-11-30 11:29:10.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-30 20:02:16.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-30 20:53:05.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-11-30 22:05:15.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-01 02:48:52.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-12-01 04:08:21.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-12-01 08:06:24.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2021-12-01 10:50:22.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2021-12-01 23:58:52.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-14 16:59:25.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-02-14 17:42:23.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-02-14 17:59:33.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-02-14 19:47:14.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-15 04:02:40.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-02-15 04:31:03.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-02-15 12:29:40.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-05-26 11:42:19.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-26 13:01:43.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-08-29 15:07:34.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-08-29 15:11:48.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-15 02:21:18.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-10-20 15:32:40.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-11-13 21:39:56.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-11-15 02:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-01-12 08:16:07.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-01-18 03:27:04.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2024-01-24 08:15:23.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2024-02-05 23:34:20.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"09dee4fe109320cb2d4a70814a9c14a0fab9f5ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-11-29 18:19:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfbc852c2bf73e0faead05a351f777b53960695e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 19:00:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/816debf5b70943ffaa22662a8faca0e8 : FAILURE in 16m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ef094d25ff545ca8dc4c883dea67e16 : FAILURE in 24m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2889e65ee76ac4232eb9cafd37cce6aa7c903128","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-29 20:25:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/68/819668/1/check/iscsi-driver-alletra-6k/401c007/401c007.html : SUCCESS in 1h 40m 48s","accounts_in_message":[],"_revision_number":1},{"id":"dfc03d0ec47567e240e0ed194d3e77115830f49e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-29 20:35:50.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819668/1/iscsi/ : FAILURE in 1h 03m 40s\n- cinder-smb http://cloudbase-ci.com/cinder/819668/1/smb/ : SUCCESS in 56m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e0c8933e40c0f535463f164370d0860195833c1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 20:46:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c611733147054d1db7ffbcdad83dd9e2 : SUCCESS in 8m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73880abae77f4e32b9934f16b0bc7edb : FAILURE in 11m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4df021b0ea474b3087adff047e68f386 : FAILURE in 12m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc80638c99954ac3a3856cc10f7b0378 : SUCCESS in 12m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/1f36b97f1fb5449e995e16af4342d0f6 : SUCCESS in 53m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dfe0e27720094ca58c708fb567f08af2 : SUCCESS in 1h 18m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/751ebc52301a498cbaee9724db323612 : SUCCESS in 7m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0a9b55d1aaca44dabaab10ec9c6db215 : FAILURE in 12m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/777d38488c1845d5b3b3122d4b047b42 : SUCCESS in 7m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5b184fccf2934a8a8fa927871f1cdbf9 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b5fd3fabfa7e496b9f54625dacebd6c7 : SUCCESS in 8m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d099212129844c698311fd9e44148629 : SUCCESS in 7m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b900474bee884ac5be1529e38c7e361b : SUCCESS in 32m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/68fec92f5c5a453f9c2c407401aceb69 : FAILURE in 7m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8dc8c33ee27e46cebe4ffe2e2af1d31e : SUCCESS in 1h 02m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cf6702cac68244139eb623dc95bdb250 : SUCCESS in 1h 37m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/206fdf2dc9744afea658bbfcedae5b50 : SUCCESS in 1h 24m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c7d0514945aa442caceeab13345e0ae6 : SUCCESS in 1h 06m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/97096ac311bc424492a050ccd8d4d906 : SUCCESS in 1h 14m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8d37bf462c6545109f3251cd88954d89 : SUCCESS in 47m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c5520b88e95c4320800a0b4ba0c6a1ee : SUCCESS in 1h 09m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f3a1a535c8bd484eb304cf093f965e08 : SUCCESS in 2h 13m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3eecc4b475a4bcfa9d51436888556fc : SUCCESS in 1h 00m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fdbe1f460c0c427b948a165d70be0888 : SUCCESS in 48m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4fd8abf358730933bd033f057f1f4d5ad2657987","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-29 21:49:24.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/c136419671f547bf9e8feaa977d9fcf7 : RETRY_LIMIT in 18s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/68/819668/1/check/devstack-plugin-ceph-dsvm-focal-py3/c2dbec4/ : SUCCESS in 2h 48m 58s","accounts_in_message":[],"_revision_number":1},{"id":"4fe8af4cc6f13813f976311bb3c3ed4e14e98888","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-29 21:56:12.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/68/819668/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"a3f1fc600006bfb33519efa6d0343658ebda2415","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-29 23:57:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/68/819668/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/076428e/ : FAILURE in 2h 42m 02s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/68/819668/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/be05398/ : SUCCESS in 1h 13m 49s","accounts_in_message":[],"_revision_number":1},{"id":"3cc92236652540afc0218babbeee7e1239e8aba0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-30 01:05:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/68/819668/1/check/iscsi-driver-nimble/f1bbd92/f1bbd92.html : FAILURE in 1h 09m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e0be0633e7711aa2d2524cce142258af90ab1141","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-11-30 01:42:30.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/68/819668/1/check/DellEMC_Unity_FC/168ba89/DellEMC_Unity_FC/206 : SUCCESS in 1h 10m 35s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/68/819668/1/check/DellEMC_Unity_iSCSI/8fbf1dc/DellEMC_Unity_iSCSI/208 : FAILURE in 2m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"6c8f8a750a82debefecc45285e09fe4f8104298c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-30 01:44:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/68/819668/1/check/dsvm-tempest-cisco-zonemanager-job/f997c89/ : SUCCESS in 1h 08m 46s","accounts_in_message":[],"_revision_number":1},{"id":"89a0fadadaed9347e99dfff48631a5c839861b7b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-30 01:56:01.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819668-1/ns5_nfs : SUCCESS in 4h 32m 39s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819668-1/ns5_iscsi : SUCCESS in 4h 36m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"228361d7fd210f3a97dfa3a3c1da923b8b9ae68c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-11-30 02:56:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4d40cac7dcc84fa7b4b75f529566a659 : SUCCESS in 1h 19m 19s","accounts_in_message":[],"_revision_number":1},{"id":"904b6bd7db236345ad4466bae80ea30b684519c6","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-30 05:13:22.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/68/819668/1/check/DellEMC_XtremIO_FC/fc19775/DellEMC_XtremIO_FC/245 : SUCCESS in 56m 57s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/68/819668/1/check/DellEMC_XtremIO_iSCSI/31cd21c/DellEMC_XtremIO_iSCSI/262 : FAILURE in 1h 54m 19s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"591d7116a24c6899b6b89b332d9022b54dcf3ac7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-30 05:42:41.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/68/819668/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"f3ff9df2936fa117abb8f6a284c3a1fe8c813880","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-11-30 07:14:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/68/819668/1/check/dsvm-tempest-TOYOU-ACS5000-driver/d50208e : SUCCESS in 2h 07m 07s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/68/819668/1/check/fc-test/1fe17ca : SUCCESS in 2h 00m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c8cc08cda89312c5be413a955de98c27656a2feb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-30 08:33:23.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/68/819668/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"cca33674cfd4e1eddf696499228c40a4f2525aa7","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-11-30 11:29:10.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/1/20211130173744/NEC-ISCSI : SUCCESS in 2h 39m 56s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/1/20211130173744/NEC-FC : SUCCESS in 2h 51m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2724f3516656ad4866b2e403589159c39f570cd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-11-30 17:25:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f0ab41a3bba153591b0db8913359c60407cb5085","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-30 17:28:44.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"d8fe53e6329f28f4887c447d40509dbba504c111","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-30 17:54:34.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819668/2/iscsi/ : FAILURE in 13m 54s\n- cinder-smb http://cloudbase-ci.com/cinder/819668/2/smb/ : FAILURE in 13m 51s","accounts_in_message":[],"_revision_number":2},{"id":"1fe312271e7ea5f93e8699ad95082d969e924e5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 18:25:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b1715135ab7401f9c80bc0ed5b3129a : SUCCESS in 19m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce473379ae4448c4b3272562dbcddd92 : SUCCESS in 20m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f29ac71884a720b0bcd7c12d9aca1692824df82f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 19:41:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f05d1848198d44788b1472eaaa570f58 : SUCCESS in 8m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a62887c8e5a5413091a8e4ef105e853a : SUCCESS in 12m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4a34ad8f3554749b77e85ae3b2c5533 : SUCCESS in 12m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff611c6509e44afe9f458b8269e13301 : SUCCESS in 16m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/147a3bf80bfc4572884342c89ed00e19 : SUCCESS in 59m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2cdfc45785424873921139e8126863bc : SUCCESS in 1h 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/737eb369ca2e4afda158b487fa3bf211 : SUCCESS in 9m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9a74e33510a24330afc77c5b3b3b811a : SUCCESS in 15m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9c2459d672384da9b7bfdd8de5421227 : SUCCESS in 7m 50s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/754ab82318c54cbfb73254b254a46d31 : SUCCESS in 9m 49s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/38854dc9a354444d8fb1e5610c2cc235 : SUCCESS in 9m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2de7ed0d1a304756a8c7a6539983fc56 : SUCCESS in 8m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f290a14eafb84e15bfb77f56ac69e438 : SUCCESS in 38m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7c19f7a723fc47faaae91d78c169d2c9 : FAILURE in 8m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aedf8cfed10a47fea0e3a34174e59507 : SUCCESS in 1h 06m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e9b09ce4067c4bc08edaff5a611b4c23 : SUCCESS in 1h 31m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/03ec33942bc844d8891d420244497501 : SUCCESS in 1h 48m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/577c1a7cf6204164866a9a1b2bb030e9 : SUCCESS in 1h 02m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bd31ac2076e340bfbc4fa631db00a8ad : FAILURE in 1h 31m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/281af18e9ef44297a1dce2c42cca8d16 : SUCCESS in 35m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/793cd636c32c432da4b9b1b000cac29b : SUCCESS in 1h 13m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/465c98c54b654185969f8789b351300d : SUCCESS in 1h 58m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efccdc6e73b7450d9fed169ef6373b6c : SUCCESS in 40m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/404f9e0a861c4c46b4ba8d8ee4bad472 : SUCCESS in 39m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f9bf71895616301fc5fafb20fd7b7561a654e743","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-11-30 20:02:16.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/68/819668/2/check/DellEMC_PowerFlex/3e3d5ad/DellEMC_PowerFlex/161 : FAILURE in 19m 57s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"63e180799af1d6eef12f2b506b93bc832de061b7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-30 20:06:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k finger://openstackcifc-virtual-machine/4fcad26360d34defb9fb1feff16d692b : RETRY_LIMIT in 10s","accounts_in_message":[],"_revision_number":2},{"id":"1ce5651bcc6e4f721f117427c00bb6dc9d3f6b94","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-11-30 20:53:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/68/819668/2/upstream-solidfire-check/cinder-solidfire-iSCSI/4a3c275/ : SUCCESS in 1h 07m 15s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"484ddebd609999853e9bed3022bd539d7b7a7938","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-30 22:02:50.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/68/819668/2/check/tempest-dsvm-full-focal-py3/4f846a0/ : RETRY_LIMIT in 47m 44s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/68/819668/2/check/devstack-plugin-ceph-dsvm-focal-py3/d8564d4/ : RETRY_LIMIT in 7m 06s","accounts_in_message":[],"_revision_number":2},{"id":"4705b8e6dc1d76503d5b562164312e0fee8ef2ad","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-30 22:05:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/68/819668/2/upstream-check/cinder-cDOT-iSCSI/398e20f/ : SUCCESS in 40m 39s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/68/819668/2/upstream-check/cinder-cDOT-NFS/a7e9357/ : SUCCESS in 37m 03s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/68/819668/2/upstream-check/cinder-cDOT-FCP/86a26ef/ : FAILURE in 2h 14m 19s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"d8461153765a1c4125061dfbbeb6feadbcd5f9d3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-30 22:25:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/68/819668/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/9da9f37/ : SUCCESS in 2h 27m 09s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/68/819668/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/bef90e8/ : SUCCESS in 1h 05m 31s","accounts_in_message":[],"_revision_number":2},{"id":"8db82e9f6262cc52f0e59513fedec8384a7b1fee","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-30 23:39:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/68/819668/2/check/iscsi-driver-nimble/ecd33d6/ecd33d6.html : FAILURE in 58m 11s","accounts_in_message":[],"_revision_number":2},{"id":"cc35d8f0b37555deef2cc0fbe5a0241372aa6a34","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-01 02:29:32.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819668-2/ns5_nfs : SUCCESS in 4h 40m 36s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819668-2/ns5_iscsi : FAILURE in 4h 44m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"54765bc5968d4b5303b5a08a665505f6a2a5706d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-12-01 02:48:52.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/68/819668/2/check/DellEMC_PowerStore_iSCSI/55b0724/DellEMC_PowerStore_iSCSI/130 : SUCCESS in 1h 33m 58s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/68/819668/2/check/DellEMC_PowerStore_FC/55da28b/DellEMC_PowerStore_FC/141 : SUCCESS in 1h 01m 10s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"83ddb65df76e8a6b7f5c2af6e9c28ebf653af9b2","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-12-01 04:08:21.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/68/819668/2/check/DellEMC_SC_FC/5f3a15c/DellEMC_SC_FC/84 : SUCCESS in 52m 31s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/68/819668/2/check/DellEMC_SC_iSCSI/0d28956/DellEMC_SC_iSCSI/83 : SUCCESS in 1h 46m 12s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"1dc05b1a3ac86c163215dfacc18d88c995ed3473","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-01 04:56:26.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/68/819668/2/check/DellEMC_Unity_FC/cca1fa7/DellEMC_Unity_FC/226 : SUCCESS in 1h 07m 29s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/68/819668/2/check/DellEMC_Unity_iSCSI/2e4c68a/DellEMC_Unity_iSCSI/228 : SUCCESS in 1h 23m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"55be9b37e6f60ca5c217f59a151ad32429bc657a","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2021-12-01 08:06:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI http://os-logs.tristero.net/hpmsa-ci/819668-2/HPMSA-iSCSI : SUCCESS in 1:12:34\n- HPMSA-FC http://os-logs.tristero.net/hpmsa-ci/819668-2/HPMSA-FC : SUCCESS in 49:50.98","accounts_in_message":[],"_revision_number":2},{"id":"b065d7ee05a25b338792d977255dd2c7aadcbfcb","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-01 10:29:11.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/68/819668/2/check/DellEMC_XtremIO_FC/22544dd/DellEMC_XtremIO_FC/263 : SUCCESS in 1h 00m 36s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/68/819668/2/check/DellEMC_XtremIO_iSCSI/862bc33/DellEMC_XtremIO_iSCSI/280 : SUCCESS in 1h 43m 16s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"f4183041491282862ad30e09092785b22de2fb13","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2021-12-01 10:50:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI http://os-logs.tristero.net/lenovo-ci/819668-2/Lenovo-iSCSI : SUCCESS in 1:28:41\n- Lenovo-FC http://os-logs.tristero.net/lenovo-ci/819668-2/Lenovo-FC : SUCCESS in 56:30.39","accounts_in_message":[],"_revision_number":2},{"id":"00ada1cbe7fea8413c1f26bef4bf110679596a23","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-01 16:38:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5f54e622b1b7461bb9b33881ecf92007 : SUCCESS in 1h 20m 38s","accounts_in_message":[],"_revision_number":2},{"id":"5f2e43491f0edb8eb40119dd93426420bd68e97c","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-12-01 23:58:52.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/68/819668/2/check/DellEMC_PowerMAX_FC/b74532f/DellEMC_PowerMAX_FC/93 : SUCCESS in 3h 12m 30s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/68/819668/2/check/DellEMC_PowerMAX_iSCSI/a986ba7/DellEMC_PowerMAX_iSCSI/91 : FAILURE in 3h 13m 29s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"1aa5a212716ca9288bb92f03a1d93d311f7da673","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-12-02 10:44:31.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/2/20211202165059/NEC-FC : SUCCESS in 2h 45m 20s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/2/20211202165059/NEC-ISCSI : SUCCESS in 2h 53m 29s","accounts_in_message":[],"_revision_number":2},{"id":"554837a4edb1767762c188568ee5fd3c9c09736d","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-02 13:38:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/68/819668/2/check/dsvm-tempest-TOYOU-ACS5000-driver/3576516 : SUCCESS in 2h 36m 45s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/68/819668/2/check/fc-test/675a126 : SUCCESS in 2h 04m 17s","accounts_in_message":[],"_revision_number":2},{"id":"279043aa6ef35308118ce5fc52f919b39b141caa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-02-14 14:25:16.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"8e1497795e981df403fe841fae2444ef326942fb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-14 14:58:05.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"73707a2a3d0e6e5ca4bf20fc4fb27d56e32d1c1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 16:08:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/07f86138ebb94f988b610589a8ef1970 : FAILURE in 21m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7383d8ee9cea4b9b8d84f9a4bd43fd64 : SUCCESS in 23m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"56628039616b7b085f203795e1fcef68e2036379","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-14 16:43:58.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819668/3/iscsi/ : SUCCESS in 50m 09s\n- cinder-smb http://cloudbase-ci.com/cinder/819668/3/smb/ : SUCCESS in 1h 11m 04s","accounts_in_message":[],"_revision_number":3},{"id":"9096f40c3feb192aafac6396ed8bca8f8ec53dec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 16:47:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7b47300aeec406cae4fa90a93ce71d5 : SUCCESS in 9m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cf4312875004e64bb9425409586c48c : SUCCESS in 9m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b4ebe5d9cea42c3a538eda165fc224d : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d853148faae40cdbc011ad78ba41316 : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/cdb4256c62c74920af5ac4926467c1bd : SUCCESS in 55m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a086d3a9184c4dc0897ef56dbb2e4cbf : SUCCESS in 1h 08m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27cb1f6c62dc4dffbf31dd625905717f : SUCCESS in 43m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17a2d7e555ca45ef9eff1e1e703c9d71 : SUCCESS in 6m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e0a37103ae794924876833105c1b0f27 : SUCCESS in 12m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2ed7809afec0404581d2456aa472334c : SUCCESS in 6m 42s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3f0684aa58ae478c81e21ebd07f1df35 : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1d7a79e6110f417680b638e3b69b0026 : SUCCESS in 7m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6e29723843c44737bd21f2ad1f880a07 : SUCCESS in 8m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d24d8c2c0e114b208c855204528e65a0 : SUCCESS in 42m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/971fea151f55412e9fceda98b5680739 : FAILURE in 6m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ef62454ebc8846ebbdb42fc19f8ed232 : SUCCESS in 1h 11m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c9c1812a25e845c09bdc3a5fd33b028a : SUCCESS in 1h 22m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9d98bc28018f4a80b6ceb9efe47c522d : SUCCESS in 1h 28m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/80fe8756568f41f7a8e353f7c31a6840 : SUCCESS in 59m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6f44d477e44e4bdebc7ef902d5a9910e : SUCCESS in 1h 16m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c85501e2f0074b569ec073c432c24c22 : FAILURE in 38m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/91441a801af0470986fb8aa00f5eb79d : SUCCESS in 1h 15m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/347c74940c844a0fad3a90c90879bfad : SUCCESS in 2h 17m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bc62274e12642fa936203f953ab9066 : SUCCESS in 44m 41s","accounts_in_message":[],"_revision_number":3},{"id":"3bc2dadcf85c55f17a11e5c1acb118aedadbcd8c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-14 16:49:11.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/68/819668/3/check/DellEMC_PowerFlex/61f10b2/DellEMC_PowerFlex/418 : FAILURE in 1m 28s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"4089de8b242d54a3eb3937aa20a8a5d7bc766244","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-02-14 16:49:42.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/819668/3/upstream-solidfire-check/cinder-solidfire-iSCSI/23260cf/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"8c6dcf1eea693f389248df72f7d7cc762a73603b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-02-14 16:59:25.000000000","message":"Patch Set 3:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/14_02_2022_16_51_39\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"85c61c057742d759822fedb2a363abdb29790708","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-14 17:42:23.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-68-819668-3/1644857230 : SUCCESS in 17m 27s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":3},{"id":"670287ec3cbfc62ff9fc0be770736f4a304d152e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-14 17:59:33.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/68/819668/3/check/iscsi-driver-alletra-6k/4a161b7/4a161b7.html : FAILURE in 49m 52s","accounts_in_message":[],"_revision_number":3},{"id":"d87abd62221c9fb864293aa957871e850f75dadc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-14 18:32:48.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/68/819668/3/check/dsvm-tempest-cisco-zonemanager-job/32c9dfb/ : SUCCESS in 1h 06m 19s","accounts_in_message":[],"_revision_number":3},{"id":"1a136923b61c988c22a96886538268e0afd57bf7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-02-14 18:46:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/68/819668/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/09ece4e/ : SUCCESS in 2h 13m 57s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/68/819668/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/bc4c7f7/ : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":3},{"id":"610452549749b05679e7c59b250ac4461698aa3d","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-02-14 19:13:47.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/68/819668/3/check/DellEMC_PowerMAX_FC/06c40a7/DellEMC_PowerMAX_FC/324 : FAILURE in 55m 35s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/68/819668/3/check/DellEMC_PowerMAX_iSCSI/1edc08a/DellEMC_PowerMAX_iSCSI/337 : FAILURE in 27m 14s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"1355c826f4cfe36a77728333fda8cf354203e72c","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-14 19:47:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/68/819668/3/ : SUCCESS in 1h 41m 05s","accounts_in_message":[],"_revision_number":3},{"id":"e4250639b4562438d48d9fb1b75e26d1894aaf75","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-14 20:25:31.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/819668/3/upstream-check/cinder-cDOT-iSCSI/6252cf6/ : SUCCESS in 35m 04s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/819668/3/upstream-check/cinder-cDOT-NFS/d7f4b14/ : SUCCESS in 33m 47s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/819668/3/upstream-check/cinder-cDOT-FCP/cb98668/ : SUCCESS in 1h 41m 36s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"9630a882377e79bca42a4194daa96ff3de31c6a7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-14 23:39:54.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819668-3/ns5_nfs : SUCCESS in 4h 40m 47s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-819668-3/ns5_iscsi : SUCCESS in 4h 46m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"4ce1643f2d161a7ec4bbf58c3fb74878cc4cd20a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-15 00:17:52.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/68/819668/3/check/DellEMC_XtremIO_FC/8bb8a6b/DellEMC_XtremIO_FC/754 : SUCCESS in 56m 15s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/68/819668/3/check/DellEMC_XtremIO_iSCSI/4adabbc/DellEMC_XtremIO_iSCSI/764 : FAILURE in 15m 00s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"cec47e251edda68e81996be942604268d7b46f03","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-15 01:26:31.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/819668-3/HPMSA-iSCSI/ : SUCCESS in 52:30.56\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/819668-3/HPMSA-FC/ : SUCCESS in 42:42.04","accounts_in_message":[],"_revision_number":3},{"id":"50d7778b191ea3fcff5ea823d47267d63c1a817b","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-15 04:02:40.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/819668-3/PVME-iSCSI/ : SUCCESS in 1:05:01\n- PVME-FC https://os-logs.seagate.com/dell-ci/819668-3/PVME-FC/ : SUCCESS in 1:02:50","accounts_in_message":[],"_revision_number":3},{"id":"7d140d52d7a3baf2150a803959ca762c9e866868","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-02-15 04:31:03.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/819668-3/Seagate-iSCSI/ : SUCCESS in 1:03:28\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/819668-3/Seagate-FC/ : SUCCESS in 1:03:11","accounts_in_message":[],"_revision_number":3},{"id":"d2e7f9b1fa5a9d5d3c313461a5629c3bc233ab0d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-02-15 12:29:40.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/3/20220215155429/NEC-V-ISCSI : SUCCESS in 5h 16m 27s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/3/20220215155429/NEC-V-FC : SUCCESS in 5h 35m 07s","accounts_in_message":[],"_revision_number":3},{"id":"7a69ed66b6af7c5ef8aac3a783e8f811e37ac882","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-15 20:53:01.000000000","message":"Patch Set 3:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/819668-3/Lenovo-iSCSI/ : FAILURE in 1:25:15\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/819668-3/Lenovo-FC/ : FAILURE in 6:01:50","accounts_in_message":[],"_revision_number":3},{"id":"872e9cb35da14285c460faad97d2bef3d1f8e6d4","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-02-16 02:58:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/3/20220216054046/NEC-ISCSI : SUCCESS in 3h 14m 22s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/3/20220216054046/NEC-FC : SUCCESS in 6h 17m 43s","accounts_in_message":[],"_revision_number":3},{"id":"96da67685b4273ed57973ca54a5490dd268f4021","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-05-26 11:40:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"331889d7b13425ae88788bcc2d2ad7608255b530","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-26 11:42:19.000000000","message":"Patch Set 4:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16079/","accounts_in_message":[],"_revision_number":4},{"id":"11da2dcf316849b33540800f284776456792b480","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-26 11:44:14.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16079/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"5bdd99b44414a4a4d816713de7fc97b89f0edddd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 12:06:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ec613485ffc4cc393eaa38396fcb1f3 : FAILURE in 20m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b58c02b538854ab0aadc42b276da722b : FAILURE in 22m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8f19afe35dce95deb6681cbb0cdbb9737e5c83c9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-26 12:56:33.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job finger://zuulserver:17979/0d073804e38a494f986da6ba52138f6b : RETRY_LIMIT in 54s","accounts_in_message":[],"_revision_number":4},{"id":"a461764a92ba938d359ecaaef490f21f468b8969","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-26 13:01:43.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1653565209036 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"13de07536bb6b573a8eacc99aeab15a64b4cb2e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 13:52:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a29f7ec7bde4caeb11d8319c41c31ce : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fc041ed43a24298973bd439142b808c : FAILURE in 10m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a589777e74c144899451b917241a3a8b : FAILURE in 10m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/220741e5d3914fa4a0cc31fe100bd158 : FAILURE in 10m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/025fd1ef8691493b9aade55c1c4f6175 : SUCCESS in 11m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/4ccada0b2d644e95ad62b392ee90793c : SUCCESS in 49m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6ad4f7d4a4f040a08e55be8e1b00ae25 : SUCCESS in 1h 24m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b4735bb1cbc43f78a2a9a453a369085 : SUCCESS in 56m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/994cd7b75080422e944dfb39dc82116d : SUCCESS in 7m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7d3ba5cc81674655a16e794f55cd8d36 : FAILURE in 12m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fcded5458e5848ceb23b65e528d481e1 : SUCCESS in 5m 12s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/076c757444fd4a15b4c5eccc76b3313c : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9e171339e6c24ba8bcd3d12108cfff30 : SUCCESS in 7m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/59b0c7abfa614934bdc243e0f4b9410e : SUCCESS in 6m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dfcc7ca277d14a2981b42ab4632711de : SUCCESS in 39m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db9e1db879264d3ea9e6db5eef53141e : FAILURE in 6m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d4152a1ab2ca48aca1566113b7f2561c : SUCCESS in 1h 15m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6fba920497124c2e82a65768439b02b8 : FAILURE in 1h 27m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/69371e0c10e34c789be7d0f4512bb7a6 : SUCCESS in 1h 44m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6f37c1ce40c046dc85dc4a0a795ef737 : SUCCESS in 1h 11m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/91396bba75fd4664a686d6754a8768a5 : SUCCESS in 1h 31m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9cd9b04a000544fca639458ee26b38c1 : SUCCESS in 56m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/220d547144ce49cdafc276e0ca09b57d : SUCCESS in 1h 10m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/624fbfdbe8744daca033bcc716422b5f : FAILURE in 2h 07m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e195a86c8e1a45ef99c61fbfd76a9c4f : SUCCESS in 54m 52s","accounts_in_message":[],"_revision_number":4},{"id":"01a6a91cbcbdf3ed3b5ff399d3f2ce84d0f39ee0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-26 14:05:42.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819668/4/iscsi/ : SUCCESS in 1h 10m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/819668/4/smb/ : SUCCESS in 1h 14m 42s","accounts_in_message":[],"_revision_number":4},{"id":"d939484fdafa276f07b87983ac5be547e50a5c3a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-26 16:05:29.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/68/819668/4/check/tempest-dsvm-full-focal-py3/d1f8665/ : SUCCESS in 1h 47m 09s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/68/819668/4/check/devstack-plugin-ceph-dsvm-focal-py3/69f5eac/ : TIMED_OUT in 3h 24m 14s","accounts_in_message":[],"_revision_number":4},{"id":"c50dd16edfffc32c6b5bc504f0be2bf63024fb8d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-26 16:16:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/68/819668/4/check/iscsi-driver-nimble/2273daf/2273daf.html : FAILURE in 59m 19s","accounts_in_message":[],"_revision_number":4},{"id":"b280bf42bedfec85258e16a2b6124ecd9e36db0b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-26 16:52:41.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-68-819668-4/ns5_nfs : SUCCESS in 3h 34m 58s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-68-819668-4/ns5_iscsi : SUCCESS in 3h 41m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"c3b3633a731f1269cd55b794f27d35b0c64ebced","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-26 17:54:04.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/68/819668/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/03ed6e9/ : TIMED_OUT in 3h 04m 09s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/68/819668/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/227c8cf/ : SUCCESS in 1h 33m 08s","accounts_in_message":[],"_revision_number":4},{"id":"b3ccfba0a27be0b997514620d1543e8c93a8b5a7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-27 04:55:37.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/4/20220527114127/NEC-V-FC : SUCCESS in 1h 48m 31s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/4/20220527114127/NEC-V-ISCSI : SUCCESS in 2h 14m 07s","accounts_in_message":[],"_revision_number":4},{"id":"354a2b0990734b6982dda7fb81293dedbc787cc1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-27 16:18:59.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7073931b679743f8832f0dc90529465d : SUCCESS in 1h 23m 26s","accounts_in_message":[],"_revision_number":4},{"id":"523732b6e6b3a0050060e1b8b0b4b8ec6e84e872","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-05-30 06:25:39.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/4/20220530090128/NEC-FC : SUCCESS in 3h 23m 46s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/4/20220530090128/NEC-ISCSI : SUCCESS in 6h 24m 08s","accounts_in_message":[],"_revision_number":4},{"id":"aae2c47d13ae4e1f90ba38fee30df89d302a77a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-29 15:07:25.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"c9dd8a2c9d9467672a4de58e77199be6d847ccaa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-29 15:07:34.000000000","message":"Patch Set 5:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":5},{"id":"f2af252ee76c9916dcfcb2bc7ce1bdbc665e8428","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-29 15:10:51.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"d825c9e9f43917c3700e6351423d8553c3be7edd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-29 15:10:52.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"b4364787ff3b2669b24515b95187a09f39a22282","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-29 15:10:55.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"03589b921a867c7a67c58cadf7eac79c1ff88dd0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-29 15:11:01.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"ac25c4ade9f8569acabb0fdd5c4dd3c72a89d89e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-08-29 15:11:11.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"57149bac8e618f014147cea6c3930b16fa58421c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 15:11:33.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"8db41691c279ae380fff05b566e5d5f5126b6ef4","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-08-29 15:11:34.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"40252504dcbf96cee045d3be66961efc9e9c3dd2","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-29 15:11:48.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"b2cb3eef04613c79d8356226def7a54b4430f706","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-29 15:11:50.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"e86746c27234f4128c0d3ce1561d497f192bd5d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 15:11:53.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"c11405876bd39b1bab5c316611a7085adb8fca38","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-29 15:11:54.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"8728e3af6ddbc2e0d3431db88fd23476bd5325a3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-29 15:11:54.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"67a9552ebc8671afb5ef98959dc7ff227c15beee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-29 15:12:07.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"98325c6e3de75d879899df51baf2c98224bddd62","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-29 15:12:18.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"4f13f96dfd5fd6313d5d7ee7325a0ba613de66e0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-29 15:14:21.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"08c54a23ddfe3e9e3b2c10db5f050d8bbdd76edf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 16:31:13.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/220cc4b82bcc4e6ca05414f5632832ab\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd7b77c0a3c94f1984c7bf3376acc8fb : FAILURE in 14m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d33e821b53dc43e0a313e3f4f54b02b9 : FAILURE in 17m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d0603064822d50d1e0fd639dd7a1f42910b412ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 17:23:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5000a79f95e44b2baf601abcc8d8406\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f49416692f384e04a541ddfba8d54200 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62e5b4e7abc442338bac1f55eae8d064 : FAILURE in 11m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bbc905190ff4c87b5fe5b8cbf6e5dff : FAILURE in 11m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7edc62dc868a47edbc2250bbf7f95256 : FAILURE in 11m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59d147f3f511400a95272609fa27890c : SUCCESS in 14m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/d1c228336b0f4c839d8d87d3314952a4 : SUCCESS in 51m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c82b226a3d5c4792a8ea66840c63fd01 : SUCCESS in 1h 12m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ff35cb0e17440e88989b0418cbd6648 : SUCCESS in 42m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ab7c59346ae4589a7ba57b3f9ffef41 : SUCCESS in 8m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6f4e6f256c0849da8c09a75eadd23247 : FAILURE in 13m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/254f5a734370496c97ea9b00f8883c37 : SUCCESS in 7m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e73c82e34ecc498f9a89a5064194c3b9 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0f14a7ea6f4f44cab67396cc2d1a2926 : SUCCESS in 6m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fc3bd0b1c4514b25963574b2ef6137e5 : SUCCESS in 10m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fb2b58c17945488da39ff5339902da57 : SUCCESS in 47m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5f70ce1767b7476a817de23aca1db08e : FAILURE in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bf7911dabfe5452c811da97c2a66ddcc : SUCCESS in 1h 07m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/81d3d09da72b448ab20e51553cd4b81f : FAILURE in 1h 35m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6d435201d9a8470a9465a35c5e4a0a54 : SUCCESS in 1h 50m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/69b10e77a8094a8b898ed09bbab97a84 : SUCCESS in 1h 10m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c427720d640741699d8383c3a61fd31d : SUCCESS in 1h 32m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7d9cf7ecf90c43c6ac23673e19101f98 : SUCCESS in 34m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/00df9bf7fb5f424ab0f7bb91a41f1791 : SUCCESS in 1h 19m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7d407bed3b7c421b8c3c72bfd7eec2bf : SUCCESS in 1h 43m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f430f0a8f17f40dba4111108df470be3 : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":6},{"id":"91a5077d7d3a2f9efbd8c782fe7fbaa8aa916895","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-29 21:37:01.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/68/819668/6/check/devstack-plugin-ceph-dsvm-focal-py3/d71e43c/ : SUCCESS in 1h 57m 19s","accounts_in_message":[],"_revision_number":6},{"id":"d2913e3a76b2afa14729dd0dbc665ba0c278f99a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-29 21:44:54.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/68/819668/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a6642ef/ : SUCCESS in 1h 35m 00s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/68/819668/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/db942ff/ : SUCCESS in 1h 45m 22s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/68/819668/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/2d8bce0/ : SUCCESS in 1h 57m 40s","accounts_in_message":[],"_revision_number":6},{"id":"b2fd20f2a7e855ba3478c2d24b21642b37180117","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-30 00:03:55.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661785658409 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"895313cb1f2daf8ea3346009a9b208d35023b10b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-30 02:11:21.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819668/6/iscsi/ : SUCCESS in 55m 26s\n- cinder-smb http://cloudbase-ci.com/cinder/819668/6/smb/ : SUCCESS in 52m 53s","accounts_in_message":[],"_revision_number":6},{"id":"703d963e8c5bd472d074a7f639abd39e0ce5dad8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-30 02:16:29.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-68-819668-5/ns5_nfs : SUCCESS in 3h 31m 7s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-68-819668-5/ns5_iscsi : SUCCESS in 3h 38m 19s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"a86ad1b5033b0e4fc297898d842d099f1d65c657","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-30 03:23:20.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/68/819668/6/check/iscsi-driver-alletra-6k/cc655a4/cc655a4.html : FAILURE in 54m 23s","accounts_in_message":[],"_revision_number":6},{"id":"0d702e8f7f8030e9cb6a2c0f5d060bbbdc976312","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-30 05:27:05.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/819668 : FAILURE in 44m 08s","accounts_in_message":[],"_revision_number":6},{"id":"841d59a8751eda18efb24c2b1dea337b6a986a6e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-30 05:43:15.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661785938447 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"5d6ebc8b22c38ac3f4a81217e98aca5fa9fe956a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-30 05:49:44.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/819668 : FAILURE in 1h 09m 29s","accounts_in_message":[],"_revision_number":6},{"id":"ef965593b20077366d7552f5c84924ca28fe41a9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-30 08:18:22.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-68-819668-6/ns5_nfs : SUCCESS in 3h 37m 25s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-68-819668-6/ns5_iscsi : SUCCESS in 3h 40m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"ce9897b7f7472d9dd227fe23b95f310ef37ecf77","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-30 08:41:03.000000000","message":"Patch Set 6:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/68/819668/6/check/fujitsu-eternus-dx-iscsi/1bb942c : SUCCESS in 1h 47m 21s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/68/819668/6/check/fujitsu-eternus-dx-fc/3c7d0d6 : SUCCESS in 1h 41m 50s","accounts_in_message":[],"_revision_number":6},{"id":"cfef7d243b1f9cf08a6a94211faf57249ebaa042","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-31 19:37:43.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/68/819668/6/check/DellEMC_Unity_FC/9a6f356/DellEMC_Unity_FC/1617 : FAILURE in 14m 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/68/819668/6/check/DellEMC_Unity_iSCSI/cfd0871/DellEMC_Unity_iSCSI/1484 : FAILURE in 2h 56m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"5d8eca88888940c1521fc09668b3628e16ae0c6a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-09-02 07:46:43.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/68/819668/6/check/dsvm-tempest-cisco-zonemanager-job/c7f22df/ : SUCCESS in 1h 09m 46s","accounts_in_message":[],"_revision_number":6},{"id":"76610652714152ca5aa6b1dc5f2c230d56df5602","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-09-05 16:17:56.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/6/20220905214407/NEC-ISCSI : SUCCESS in 3h 33m 25s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/6/20220905214407/NEC-FC : SUCCESS in 3h 33m 45s","accounts_in_message":[],"_revision_number":6},{"id":"bdd9c518521aef09ca6a81898954e4302ae9341c","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-09-07 19:51:24.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/68/819668/6/ : SUCCESS in 1h 32m 16s","accounts_in_message":[],"_revision_number":6},{"id":"dd164b8965529a49cb907db004e3f706ae0c4f9d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-09-13 20:02:06.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/6/20220913221357/NEC-V-FC : SUCCESS in 6h 04m 47s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/6/20220913221357/NEC-V-ISCSI : SUCCESS in 6h 48m 05s","accounts_in_message":[],"_revision_number":6},{"id":"f01f48c8fe388e4d077703d1294ab1f41eddad7a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-09-16 10:01:17.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/80f0831593a6443a8b5309aad9a8a4d8 : DISK_FULL in 2h 58m 09s","accounts_in_message":[],"_revision_number":6},{"id":"ec15e58f7f54062a0de0e3c7dd8f97b08c7ef37a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-03-14 18:08:06.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"bc277aa1bd1772b1c08fb267073758d2b3cadc7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 18:38:44.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14383787b6c144ac8a4069993a82f469\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e428b448d9149dc8b5e0cb3ec9f419f : FAILURE in 26m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a1b4a482c4f144d1b493ff600fe92e0e : FAILURE in 27m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ab032bc2979328746bfa4c28fe477494f1028e41","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-14 19:00:50.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ecb70b059e4448ac9fe8db1fee8008ef : RETRY_LIMIT in 6m 20s","accounts_in_message":[],"_revision_number":7},{"id":"c209db0c0013659fa2bf83c92737caa8b78e8354","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-14 19:20:23.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder iscsi-driver-primera-cinder : RETRY_LIMIT in 0s","accounts_in_message":[],"_revision_number":7},{"id":"941bccbbcb509c9208af11a759fd515beea862fb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-14 19:44:35.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/819668 : FAILURE in 26m 56s","accounts_in_message":[],"_revision_number":7},{"id":"18127211af3c8a7289b649f42036b80b2bea008e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 20:30:14.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f3bcb3da7ea4a28a5012d2db811b066\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ea4758988134517b5d8d9ef3e6e525d : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37bc000db00548429e458dcd8555f260 : FAILURE in 9m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15bf18a25e104799b0f813d0ddbdb566 : FAILURE in 10m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1a2db11e8654b83bcdeec227dd7afe5 : SUCCESS in 14m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/f991d495d5d04191aa1ab3c195993faa : SUCCESS in 56m 34s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/86ec73777ffc490c8929b255ede7550c : SUCCESS in 59m 23s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/67278f3fca684925b394d227f043da67 : SUCCESS in 1h 28m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/faee5ceb477146ca905cadb9a227c246 : SUCCESS in 55m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/274470e0a0ea446ca4ac72e2945efad9 : SUCCESS in 8m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/231a6ac4519f45b79581da9d4197109c : FAILURE in 13m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/af893cc1eee54374be32cda5becf6278 : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/62449c755f2b4994b53c4b36547db5c2 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8cdd18891bf2453490b3d76c05d00ab5 : SUCCESS in 5m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ba3a276eabd94fc585c4749fae6224ae : SUCCESS in 5m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e3ba59a224bb4f56aedd1fd2343003b5 : SUCCESS in 47m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a6e49d8154ae433b8667284222c4aa89 : FAILURE in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/92b35a3485044183aeea88a1e8f5cc29 : SUCCESS in 1h 20m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/57a3dcd5eb404118a9072e618c4e45d6 : FAILURE in 1h 18m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/49e5c7c284c04d30a4200cc666a88081 : SUCCESS in 1h 43m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f71f871dad6b44a8bcd0cb310fc8ea37 : SUCCESS in 1h 07m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/535fc1a175d646b1924570e31e687709 : SUCCESS in 1h 27m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/07edb59a3ba741afbe149c59ea7f1e46 : SUCCESS in 39m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c1998ee887f8445ab50fed725825e4c7 : SUCCESS in 1h 24m 13s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/59e94f0efe684c32815323545f6c2d89 : SUCCESS in 2h 13m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ba472911ff441e585125231c87440e6 : SUCCESS in 50m 20s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/808927c8ff9145db94540e82b83d2ca5 : SUCCESS in 1h 25m 20s","accounts_in_message":[],"_revision_number":7},{"id":"efebcbdbb54a25f46b65388bd86fea60f25beed2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-14 20:35:10.000000000","message":"Patch Set 7:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/102/ (1/2)","accounts_in_message":[],"_revision_number":7},{"id":"b58b009cecdae707ba052122b0e886837b6c15a8","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-14 21:42:11.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/68/819668/7/check/fujitsu-eternus-dx-iscsi/11163d5 : SUCCESS in 1h 52m 29s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/68/819668/7/check/fujitsu-eternus-dx-fc/e23cfa2 : FAILURE in 2m 34s","accounts_in_message":[],"_revision_number":7},{"id":"57036b0370ac236526670c90aa0fdbb774ed546d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-14 23:34:36.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-68-819668-7/ns5_nfs : SUCCESS in 3h 29m 49s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-68-819668-7/ns5_iscsi : SUCCESS in 3h 33m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"08c87ad85fdedcf02c14edb01205ac0c7d07a7e6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-15 00:43:19.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/68/819668/7/check/iscsi-driver-alletra-6k/1308911/1308911.html : FAILURE in 2h 10m 45s","accounts_in_message":[],"_revision_number":7},{"id":"56ff4e64718ea4ac12e5cb0023b9db386a8204f3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-15 00:51:44.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/68/819668/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/dc95a54/ : SUCCESS in 1h 43m 38s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/68/819668/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d72a3a3/ : FAILURE in 2h 05m 40s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/68/819668/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/54bc1e7/ : SUCCESS in 1h 55m 11s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/68/819668/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/aece29c/ : SUCCESS in 1h 46m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6554c6bb12b759eee544e2d61220fe55711cfda6","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-15 02:21:18.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/9f7bccd96bfc4d649e61c1f3777d4c09 : SUCCESS in 1h 09m 43s","accounts_in_message":[],"_revision_number":7},{"id":"ac089eaa2c50baa6274594d10f0bfe8b19edb795","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-15 02:25:33.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/4c8c4b984add48cd90e331118e85d046 : POST_FAILURE in 2h 06m 53s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/4d0df36469bb497dbf7bdc02a7241d87 : POST_FAILURE in 1h 52m 44s","accounts_in_message":[],"_revision_number":7},{"id":"5671fde83703fe1289ddcdf04119b5efd3b8ee40","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-15 09:39:14.000000000","message":"Patch Set 7:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1177","accounts_in_message":[],"_revision_number":7},{"id":"456a2fb552f87b4b124eac9d46d9be829804d8bd","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-03-18 14:52:35.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/7/20230318174739/NEC-V-ISCSI : SUCCESS in 2h 45m 05s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/7/20230318174739/NEC-V-FC : SUCCESS in 6h 04m 53s","accounts_in_message":[],"_revision_number":7},{"id":"07d4846e140dbf90ec2f37a1771239c71ff0803f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-03-25 18:50:19.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/7/20230325203358/NEC-FC : SUCCESS in 3h 07m 52s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/7/20230325203358/NEC-ISCSI : SUCCESS in 7h 16m 18s","accounts_in_message":[],"_revision_number":7},{"id":"7246b56940970f0c6fe49de0929bf8406a639b68","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-23 11:13:42.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/68/819668/7/check/iscsi-driver-nimble/1cc1bf7/1cc1bf7.html : FAILURE in 42m 53s","accounts_in_message":[],"_revision_number":7},{"id":"2f561800290d44d9edac07c2c517c865d297244a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-10-20 08:16:15.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"a4ae330bf77a68fe02c498fb4bfb69469412f38e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-10-20 08:22:39.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c28818533eea4cd7a198170ab13e2411 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"0b1ad34b46299e52fce54c8cd449b7fb7866d9ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 09:57:00.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7572f4e76064ac2b827304e049ea386\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4795ffc137d944f0bc1f787da73bcc9d : FAILURE in 25m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1c0fa6db5c804c759e27baf81f0806e7 : FAILURE in 20m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/12410050258b4675ba9baefa5cfce48a : FAILURE in 25m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c508f4b585ab4e64b3f1b512b4a2b19ee744ce47","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-20 10:05:11.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy pure-devstack-cinder-tempest-iscsi-aio-jammy : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/c079c0f66b4b4abc99ee86315c7d1368 : RETRY_LIMIT in 56s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy pure-devstack-cinder-tempest-nvme-tcp-aio-jammy : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"ce4e5ad6b088fb3a2b170b7cb480ac06a9e26d86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 10:45:27.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/188edcd532f74a078ed47cd0bdeb9677\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83b53d5211c242f7b1dbf8a4c9fd6e19 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8a7ec78bbc0417d9c2cd2099a6916f7 : FAILURE in 10m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b9594c4ed3e446581c34990e22c7c39 : FAILURE in 10m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65bfc01ab38542d18eb9ec0f4e697ec1 : SUCCESS in 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/20bb8ddb1d35417fb050f3ffafc33d9c : SUCCESS in 58m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/17b926c5f109462298b914046f1ac4f9 : SUCCESS in 41m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e728bb26e3b04d698cf523759d2b1140 : SUCCESS in 1h 33m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14e34ffabd76495283ed4e509cb1cd26 : SUCCESS in 1h 04m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cdc703e026648f5bcf00585243234a4 : SUCCESS in 8m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a02eca85ff3643d38aefd8737d9814c2 : FAILURE in 12m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/957bc7704100445991ab0edc482090c8 : SUCCESS in 5m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d53efb162cd741709bc050288eeb0a5a : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/455a68fa5da8404ea23846e1654f6a94 : SUCCESS in 7m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4c43aabd17f9417bb5317b56e5919d77 : SUCCESS in 6m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/04b59638a91c4ca48248b19663813d35 : SUCCESS in 49m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ed73de81a2564800aad30586c60755be : SUCCESS in 6m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/342fcbb458f14611a7a5c82bd390075a : SUCCESS in 2h 20m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/76a3326336344fd79a06594700d2e16a : FAILURE in 2h 09m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8b7d3ee677b74033a78023a47d699030 : SUCCESS in 1h 54m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/35483cd5f5b0420abbbc4dbd46e6c9b8 : SUCCESS in 2h 20m 38s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/578b3798987d4e4987a1c4bac8cbf12d : SUCCESS in 1h 13m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/afd9cffadd7c4f3bb6204e07e49a5467 : SUCCESS in 1h 37m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1c6c3dba8b0a4535b7f76bc15f4df861 : SUCCESS in 40m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e3d4ebc919844719a617ef210c911e33 : SUCCESS in 1h 30m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a8376e6b0a3d4bde8a787ea397235ca2 : SUCCESS in 1h 30m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b65454a6479c4d09ace74ccbea054e7e : SUCCESS in 1h 31m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b803cdb070b478baffa486f62f91ac3 : SUCCESS in 1h 09m 17s","accounts_in_message":[],"_revision_number":8},{"id":"d8124c530df651e4e8e285fe2fee3b67293f79c1","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-10-20 10:58:47.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/68/819668/8/ : FAILURE in 1h 10m 21s","accounts_in_message":[],"_revision_number":8},{"id":"167a45e984f159cc3c204698986c5b0d7fbfcb56","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-10-20 12:29:28.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1997/","accounts_in_message":[],"_revision_number":8},{"id":"e7c693c99325bee0f317c1a5db7292098cd56ef3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-10-20 12:38:20.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b8f3a72491e046dd83c19431a7b27d20\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0d49f947bdfd49b794f8dc4e129f7ab0 : FAILURE in 1h 22m 52s","accounts_in_message":[],"_revision_number":8},{"id":"56bfc1d8ff9df231fdc5c9dd327c0a9a8698adda","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-10-20 14:22:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/799/","accounts_in_message":[],"_revision_number":8},{"id":"2830e1952d777080b2b1cbf242a9518b70eb76fd","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-20 15:32:40.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-68-819668-8/FC/ : UNSTABLE in 2m 18s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-68-819668-8/iSCSI/ : FAILURE in 4h 46m 3s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"caab8b41a58c24b1eb0cc87202bfd6e216eb90b7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-20 16:55:21.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/68/819668/8/check/DellEMC_PowerFlex/46200ed/DellEMC_PowerFlex/781 : SUCCESS in 1h 29m 28s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"c32154c6c8c28c79641942ee21c38c58a3bd6d71","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-20 22:05:33.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/68/819668/8/check/DellEMC_Unity_FC/578d686/DellEMC_Unity_FC/907 : FAILURE in 2h 20m 36s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/68/819668/8/check/DellEMC_Unity_iSCSI/76ea153/DellEMC_Unity_iSCSI/915 : FAILURE in 2h 13m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"31e6a718f4732dd8052c44737e59bcfeb42feb64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-11-13 16:46:14.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"7f7fb126a5b6f4364c0c42ecc25a12f8b7e86708","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 17:15:08.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee8f9bd2e5d744b08eb5b2e051ee488a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd4233bdf4a249f69560c70038e397cd : SUCCESS in 25m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0fa71b1c7db14c67b03b6f5af18f45f4 : SUCCESS in 14m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/72356b488c3647c3a51877709be0b002 : SUCCESS in 25m 49s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1db2bbea47dd35ebc9326c426af797b61fc46356","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-13 17:33:49.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/489e47c3f61d4d6b96d4b3d19c1f6f91 : RETRY_LIMIT in 5m 33s","accounts_in_message":[],"_revision_number":9},{"id":"b2a7442a139f6d987eef9420e7101fe6b0deee79","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-13 18:27:11.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/62/.","accounts_in_message":[],"_revision_number":9},{"id":"0fe14f7388da426d0a337ed6d5d1a86a9070dad1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 19:16:57.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9703bff149db46b29c172ed699920805\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ae5d385731948369cff2fe66c2df81f : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e00f7cae57a45da99d62a05430955fc : SUCCESS in 11m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ff2c79e643b4bcfbdba61bed3e70f04 : SUCCESS in 13m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ba9ed859e1d44faa249129c5fda5213 : SUCCESS in 14m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/98cbdcb8776041578d4c54dbf9368fb3 : SUCCESS in 56m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/597bbe251c8f42c1b24c807f3ad7981e : SUCCESS in 50m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/35d00a1e6f864206abc12e2a865c25b9 : SUCCESS in 1h 25m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4832621e14d4fb5b107fc63a7eb8dd6 : SUCCESS in 1h 03m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95d19f356c8b4affbd4dd12a9cb35cc2 : SUCCESS in 8m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6d0345fa2ee54f6993a89e1ef36bbd93 : SUCCESS in 17m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/466c00ef2c9a4803beb02e4b23add990 : SUCCESS in 5m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a8ea3ca4b3f5455bb299760542f6c254 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/28175a519b074b4e986ad1434bd908bc : SUCCESS in 9m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1262a75a2236499cb657498960dd91cb : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ebd90000c10b46cd9c2e24b15b447c49 : SUCCESS in 56m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0fe6070569694bb28306cd59bf2a43b4 : SUCCESS in 5m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8b99bb707eba4092b22c2a83b123f74b : SUCCESS in 2h 17m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e43b98e5d0ea4c9592bfe2391899a363 : FAILURE in 2h 23m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/395614ad5fb8441c860ab9dc931fb6a3 : SUCCESS in 1h 44m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e4ed1bba855f48af8f81a0bd010556b0 : SUCCESS in 1h 51m 04s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5d74d5e7900944139455cb3eba26d5c1 : SUCCESS in 1h 24m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5650a18f937c4ca5b5052b6a4b4d4a79 : FAILURE in 1h 42m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/845b1ff7335a4ae0bc379b09498d6e17 : SUCCESS in 47m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9f30d1aa6f9c4573a608d539a7c427b8 : SUCCESS in 1h 15m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/57b18010356d4bb6833586b42b59fce2 : SUCCESS in 1h 57m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/20a8f6dfc410495bbb5b1fb6ad4889cc : SUCCESS in 1h 37m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f81e76ddbd5a4e74b4a71019426d4f29 : SUCCESS in 48m 27s","accounts_in_message":[],"_revision_number":9},{"id":"421ae5f2adb6406c44ddf9473563715042823f34","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-13 19:18:23.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/95b9ea347b53466b95fcc11a80debcd2 : RETRY_LIMIT in 56s\n- pure-devstack-cinder-tempest-fc-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/6057e5d3b62045ab92f484e8bf2b47cf : RETRY_LIMIT in 57s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/e34d2a1e17bf466180c54d3f7fe38d66 : RETRY_LIMIT in 1m 01s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/db804e3c73484fdca003bddfa4af1a40 : RETRY_LIMIT in 58s","accounts_in_message":[],"_revision_number":9},{"id":"7c5cbb9e4ea49684d7a8b825ab058406feb2c7a9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-11-13 21:10:44.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/74465a54e65a41d29a390607e288e1e3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/28d22647d2124ef7bc9e3c5aa993a8b2 : FAILURE in 1h 22m 22s","accounts_in_message":[],"_revision_number":9},{"id":"7effa1a64d2f0c3c48686aec5590a630f7abeccd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-11-13 21:39:56.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/819668/9/upstream-check/cinder-tempest-plugin-ontap-fcp/0f849b4/ : SUCCESS in 2h 22m 01s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/819668/9/upstream-check/cinder-tempest-plugin-ontap-iscsi/772e892/ : RETRY_LIMIT in 9m 01s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/819668/9/upstream-check/cinder-tempest-plugin-ontap-nfs/e35586b/ : RETRY_LIMIT in 7m 49s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/819668/9/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/8968d53/ : RETRY_LIMIT in 8m 16s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"81a4cde42b1b15eaa8e0df803a61828b79c8be3b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-14 00:23:33.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-68-819668-9/FC/ : UNSTABLE in 2m 26s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-68-819668-9/iSCSI/ : FAILURE in 5h 36m 51s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"939d9f071d9fe1007da33510295c7dbbef19a17b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-14 00:29:23.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/62/.","accounts_in_message":[],"_revision_number":9},{"id":"d2425c6ac403d03a8f572e22d8f96b671f421d91","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-11-14 11:34:48.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/819668-9/HPMSA-iSCSI/ : SUCCESS in 1:56:15\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/819668-9/HPMSA-FC/ : FAILURE in 1:46:05","accounts_in_message":[],"_revision_number":9},{"id":"0bc53c007e8e7f8172ddaa7808f13552bd8a27e6","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-11-14 14:35:22.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/9/20231114202845/NEC-V-ISCSI : SUCCESS in 3h 03m 34s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/9/20231114202845/NEC-V-FC : SUCCESS in 3h 06m 33s","accounts_in_message":[],"_revision_number":9},{"id":"429abca2cf0f061a373477b1c008b5275c7e1dc0","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-11-14 15:29:32.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/68/819668/9/ : SUCCESS in 1h 20m 13s","accounts_in_message":[],"_revision_number":9},{"id":"29bbea6fdaa66b5f5591e5711a7f2ce3b9a3c915","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-11-14 16:28:52.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/819668-9/PVME-iSCSI/ : SUCCESS in 2:27:33\n- PVME-FC https://os-logs.seagate.com/dell-ci/819668-9/PVME-FC/ : SUCCESS in 1:52:10","accounts_in_message":[],"_revision_number":9},{"id":"13a2c2343834ed809fd2f30e10adcead3603f00d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-11-14 22:39:23.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/9/20231115034928/NEC-ISCSI : SUCCESS in 3h 47m 06s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/9/20231115034928/NEC-FC : SUCCESS in 3h 49m 52s","accounts_in_message":[],"_revision_number":9},{"id":"f5026bc5a744d47b01905d7d023de053ec1bb60c","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-11-15 02:00:07.000000000","message":"Patch Set 9:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/68/819668/9/fc : SUCCESS in 0h 34m 41s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/68/819668/9/iscsi : SUCCESS in 1h 04m 44s","accounts_in_message":[],"_revision_number":9},{"id":"daa8d986b62048f620872e7c880fcbbfe82e00e7","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-11-15 17:21:27.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/819668-9/Lenovo-iSCSI/ : SUCCESS in 2:28:35\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/819668-9/Lenovo-FC/ : SUCCESS in 1:51:11","accounts_in_message":[],"_revision_number":9},{"id":"b5b81c9505652b93c544e6eba7a1f0d04bb1adfa","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-01-12 08:16:07.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e5ef3f4285e516da29902227acc431fd2156d7b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-01-12 13:13:24.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"fb05ae1d2d2c6c5ef0fedd0ab05c7c38f4ff9908","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-01-12 13:13:53.000000000","message":"Patch Set 10:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16360/","accounts_in_message":[],"_revision_number":10},{"id":"d51432c552a77ccc3b6a28ebe2bc998fea901a81","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-01-12 13:13:55.000000000","message":"Patch Set 10:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16360/ : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"1a3fb044880aa62e1bbf3d1167e63df973257e67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-01-12 13:26:05.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"9e16213ffde08a5685e1a11f0a8cd8df5fb0978c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-01-12 13:30:39.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c7ad7bb940864b3dfb675ff21a37f4d3cf7c0cf3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 14:42:03.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/753e7aee7e484f8a8e8048acb651844c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fdeeca7d2ae433cbc491cf08e16e68f : TIMED_OUT in 1h 01m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/860a370887b349f9b1cbf65a08406d75 : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/acce7175773640aa88d5060b7d607548 : SUCCESS in 10m 12s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"14cc7d8c6ecc9669faca3dfda57dd51bbbfdade6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 16:45:52.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2459cc9f99a04e8abd5b8b6b883d1ed8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3de3f4e995034cfcba9788c53a3581a6 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e7be35611d44770b5bc38899ffa89e7 : SUCCESS in 13m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/57ec7f8ba4534efb83ae01abb4450a04 : SUCCESS in 14m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29cb5bd2d8c942f49f0574ada554a30b : SUCCESS in 12m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/aacd10ba38774474b9fa60ab470b4d22 : SUCCESS in 1h 09m 11s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/33165e0f51d64b95b73476841e8ffabe : SUCCESS in 1h 00m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e305b0dcf8704002a786406d1af7b343 : SUCCESS in 1h 18m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c8f1ebcd9084cf6af4512e6cdb0565a : SUCCESS in 34m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82a94322c033477cbeb51cb789d542a6 : SUCCESS in 8m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4121f2e3d34d4d7fa0b9c8519862116d : SUCCESS in 18m 03s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/19942d3294a34e46ba6a67be032f553a : FAILURE in 13m 37s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ebe5166509b7408d85140b85991d78f9 : SUCCESS in 4m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3368fa7e5d2d4795a77f86b36928f4a0 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0157e84e439b475c911308bfff3553fc : SUCCESS in 6m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/959512331dab40f7814bbf60b3fe98e7 : SUCCESS in 6m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8999472e9b5341f6b78aaf1489a967a8 : SUCCESS in 47m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9749946c16f14c63b01f2e3019cb62c2 : SUCCESS in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d7b91ab261db45fe8eae71ab9c9ebaf6 : SUCCESS in 2h 10m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4f5644d41ba54282ae9e3d488917fa9d : TIMED_OUT in 3h 05m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1b607455848344ffa130dea9af40aba7 : SUCCESS in 1h 37m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e61db37394ea492c9fcd409111df758e : SUCCESS in 2h 09m 16s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4fd051f285694b0e88abb62d6e6cc27c : SUCCESS in 1h 16m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cdebd88e141443159f11e1603812b20f : FAILURE in 1h 32m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/75f80a748db041f38772af65d42d8f87 : SUCCESS in 57m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/63d458a17659403fa206482122e14a1e : SUCCESS in 1h 24m 33s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ac3d914a07c84078a3aa8014045bbd29 : SUCCESS in 1h 24m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/47a04c082cee4e728fea6674ebe6945a : SUCCESS in 1h 30m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d983d2a744c40ffbdac928208d8d6f1 : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":11},{"id":"c66e2200ca650422db820d09a1df3f7949b663c3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-01-12 19:14:42.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/67d64fbc580a41faa32f2d58b18d5b9d : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/16c69824d5cf4fd085ec3a640de3c088 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":11},{"id":"0b37a5af3b1637c8b00d9a98bb7692a132294e7f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-01-12 20:19:06.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/722ec814eeb145e1a2d1f9935ffef290\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c7b0d31e6d9e49c797c6ec6443ac28b7 : FAILURE in 1h 56m 59s","accounts_in_message":[],"_revision_number":11},{"id":"a41732f1c314849985a41845127aa7bf7687dcec","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-01-12 20:24:34.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/819668 : FAILURE in 49m 10s","accounts_in_message":[],"_revision_number":11},{"id":"3720e34ae094b2993ae02b90b966c448d6375b39","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-01-12 22:00:13.000000000","message":"Patch Set 11:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/68/819668/11/fc : SUCCESS in 0h 54m 11s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/68/819668/11/iscsi : SUCCESS in 1h 09m 25s","accounts_in_message":[],"_revision_number":11},{"id":"939359f5e6c38960d587799caeccbefebf13f7c2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-13 03:27:41.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/68/819668/11/check/DellEMC_PowerFlex_3.6/c98f43a/DellEMC_PowerFlex_3.6/1199 : SUCCESS in 1h 29m 46s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/68/819668/11/check/DellEMC_PowerFlex_4.5/6ff798a/DellEMC_PowerFlex_4.5/171 : SUCCESS in 1h 44m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"599c71f9177b8bd7e89816dd22e67951983d5274","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2024-01-13 11:15:05.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/68/819668/11/ : SUCCESS in 1h 35m 36s","accounts_in_message":[],"_revision_number":11},{"id":"e33fa2a80875be189476534dbe5058b3702d277b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-01-14 10:15:36.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/11/20240114151831/NEC-V-FC : SUCCESS in 3h 23m 10s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/11/20240114151831/NEC-V-ISCSI : SUCCESS in 3h 57m 02s","accounts_in_message":[],"_revision_number":11},{"id":"95e088c2fd78a443b2c1511310acbc3e11c8d915","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-01-14 14:57:29.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/11/20240114142515/NEC-FC : SUCCESS in 8h 45m 57s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/11/20240114142515/NEC-ISCSI : SUCCESS in 9h 32m 11s","accounts_in_message":[],"_revision_number":11},{"id":"2ca38a61ec60ab359ec6c595e3e586143dc52184","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-01-15 21:43:35.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/90d5692757c54760a4f4f0c89e0a3c79 : SUCCESS in 1h 08m 20s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fa248b7691be460bb4aa8d44d22bd348 : SUCCESS in 1h 08m 36s","accounts_in_message":[],"_revision_number":11},{"id":"4eba6416e8add573d2e31258ba422583213c1f31","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-01-17 17:41:08.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a018bc71b8b0ed747fdc502f940ffba909d1c8a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:03:13.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/054f924cb71643da9a6eef79e9104482\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/18dd04377cf34619b92215d465f4b2f8 : SUCCESS in 10m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b9a76377cc704794bcc1d58d730fb449 : SUCCESS in 19m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/34486608fbda4e329f7ad465218d85bd : SUCCESS in 9m 04s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"07e02785d12e84b9aeff98df9ca3e5eee2a800c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-01-17 18:10:12.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"c3291a86712e08938579e1132d62e52565ebf953","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-01-17 18:10:27.000000000","message":"Patch Set 12:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16384/","accounts_in_message":[],"_revision_number":12},{"id":"967117ef8f0842665815080076275e20a5e814e7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-01-17 18:10:28.000000000","message":"Patch Set 12:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16384/ : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"0ab90553ea8f79898c8f72147636e5a857679ecb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-01-17 19:02:16.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/af3da18b5dfe4145b8974981c5d02ca3 : RETRY_LIMIT in 7m 20s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e2270a2ae2ea40e5ac34d23024bdc21c : RETRY_LIMIT in 7m 02s","accounts_in_message":[],"_revision_number":12},{"id":"d22aad794b21c946c91010d04dbd175ce46d85dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 19:12:17.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a3e9bf95ef6479dba92e7307aa9f823\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ca7036deb46423bae3084a6d021aaf2 : TIMED_OUT in 1h 00m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e5efa4704edf4829a0da812ea22d2762 : SUCCESS in 21m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/01d200873807416ea0a40b8967fa0b1b : SUCCESS in 12m 14s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"84d4ec59cef33f9b1cacba00a033e84b6aa48137","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-01-17 20:02:26.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/819668 : FAILURE in 43m 16s","accounts_in_message":[],"_revision_number":12},{"id":"0678cd60da42fba8e115e634b3c9134b7b696609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 20:27:41.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7190543a00bb4e278a2184e0a6129220\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b1b29be9f0c4fdcb45835af781c93e2 : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6b8a077a25e47b6a62cf783558a6a83 : SUCCESS in 11m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fea4d029644449b5812daa3a8686339f : SUCCESS in 13m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45dbc1886b034e11842e948dff6e78f9 : SUCCESS in 14m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/b36432f66cb14b6091391bbb22ff8224 : SUCCESS in 58m 56s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/abdd69a3e58b454d9b58ae718a2ebaaa : SUCCESS in 1h 24m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c70829f733f0466292fc1fcf5b263aa1 : SUCCESS in 1h 33m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea38182a7e194d65ad350b445caaebfb : SUCCESS in 32m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2dcde21513854a28b6c293fa30d999f1 : SUCCESS in 8m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/473bc95634034249befa8912b8a2ce99 : SUCCESS in 17m 38s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a912059855f648bf9b90ed39281bdc9f : SUCCESS in 18m 15s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/83d0e99e52c346c2af80d1c60466bce4 : SUCCESS in 5m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/53b1360f28864d1a82f37ad5acd84d98 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/146c5f2ec6cb42209222b8429a17c04e : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2525fd753f044e2f965dc66e47525966 : SUCCESS in 6m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/93f44a548e274c98adc9d10df791fe00 : SUCCESS in 55m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/264636b1c63a4784a07fb3add5792089 : SUCCESS in 6m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2dace796bdf848388fffb447b72893a0 : SUCCESS in 2h 01m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2a821722ac7d40c1ad19b6cb8e1552b0 : SUCCESS in 2h 12m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8bf670c485c0421ab7161bf4644c997d : SUCCESS in 1h 55m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7d34c28ee9f046cc93ca9c52654b182d : SUCCESS in 1h 50m 35s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/37c8f8b359e64e7bb2514d378ab1f302 : SUCCESS in 1h 07m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f3f7f3105b6f401e851d0aa68f517d50 : SUCCESS in 1h 33m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4e1f095e56164e3fa2c23959a2d2bc6a : FAILURE in 36m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/77474482f5684d16b039f20225a97618 : SUCCESS in 1h 17m 16s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/34ecf6cfc8394451a5981dfc42a62220 : SUCCESS in 1h 48m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5a79e102e40e4751b2854b3e8f96cd14 : SUCCESS in 2h 01m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c21e273c9874cfea4d2b094574d5797 : SUCCESS in 57m 30s","accounts_in_message":[],"_revision_number":12},{"id":"2085dc091df81153856aaceee30798f1df61a44c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-01-17 21:16:04.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/7881d44e9dbb49259f20666b981900b5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/20fd6a18067f4442956b56d9ad036a81 : FAILURE in 1h 24m 46s","accounts_in_message":[],"_revision_number":12},{"id":"234d6216eab962037f70b8fbf6bfb8337b7fe106","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-01-17 22:30:29.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/68/819668/12/check/DellEMC_PowerStore_iSCSI/28e51e5/DellEMC_PowerStore_iSCSI/948 : SUCCESS in 1h 11m 30s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/68/819668/12/check/DellEMC_PowerStore_FC/caec3f8/DellEMC_PowerStore_FC/1104 : SUCCESS in 1h 23m 01s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/68/819668/12/check/DellEMC_PowerStore_NFS/e9e994f/DellEMC_PowerStore_NFS/634 : SUCCESS in 1h 04m 46s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/68/819668/12/check/DellEMC_PowerStore_NVMeTCP/5e111da/DellEMC_PowerStore_NVMeTCP/500 : SUCCESS in 1h 55m 05s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":12},{"id":"62e2bbe0c472fa003ef20aec932b73b1c728f937","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-01-17 22:36:50.000000000","message":"Patch Set 12:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/68/819668/12/check/fujitsu-eternus-dx-iscsi/ea0079a : SUCCESS in 2h 07m 36s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/68/819668/12/check/fujitsu-eternus-dx-fc/9159c86 : SUCCESS in 2h 14m 58s","accounts_in_message":[],"_revision_number":12},{"id":"d43e22975304ee6de04a0fd7755df39684a5677d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-01-18 01:06:40.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/819668/12/upstream-check/cinder-tempest-plugin-ontap-fcp/5b4756f/ : FAILURE in 1h 47m 22s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/819668/12/upstream-check/cinder-tempest-plugin-ontap-iscsi/fff8b38/ : SUCCESS in 2h 24m 44s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/819668/12/upstream-check/cinder-tempest-plugin-ontap-nfs/cbe2541/ : SUCCESS in 4h 31m 32s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/819668/12/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/e298b7b/ : SUCCESS in 2h 27m 46s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"910518cbe9b4ce6267c92f96d6ed09917c4aa6c1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-18 02:20:49.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/68/819668/12/check/DellEMC_PowerFlex_3.6/fb78876/DellEMC_PowerFlex_3.6/1234 : SUCCESS in 1h 16m 08s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/68/819668/12/check/DellEMC_PowerFlex_4.5/129c1f2/DellEMC_PowerFlex_4.5/205 : SUCCESS in 2h 01m 00s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"9a9a702c0d7939e579140539a46a159ff82cfc72","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2024-01-18 03:27:04.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"72b9341e0408e16d46dbf0c189568d42a9d35e2a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-18 04:21:01.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/437/.","accounts_in_message":[],"_revision_number":12},{"id":"cec3bb8abeecee72ff4ddcaac533da238ba1f9c8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-18 04:27:35.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/451/.","accounts_in_message":[],"_revision_number":12},{"id":"b3cdc1de73293fdfa92be7e82bbb96bcde4981af","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-01-18 13:50:29.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"e133459dbff7baf0fc872cd9cfe5e0feca4b88de","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-01-19 20:55:31.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/12/20240120022439/NEC-V-ISCSI : SUCCESS in 3h 24m 12s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/12/20240120022439/NEC-V-FC : SUCCESS in 3h 30m 49s","accounts_in_message":[],"_revision_number":12},{"id":"4e918f59a0511303ec3b9732f25176470b41ae27","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-01-20 08:07:21.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/12/20240120084309/NEC-ISCSI : SUCCESS in 4h 02m 57s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/819668/12/20240120084309/NEC-FC : SUCCESS in 8h 24m 09s","accounts_in_message":[],"_revision_number":12},{"id":"765a37c58fe460849f2c4bd40a55c73482a47b7e","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2024-01-24 08:15:23.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"36f110e3d6e1dc8279eef9e4bbb761c2d1ea10e4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-05 23:34:20.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"db3853900c151e3422794e6ba81c1c342f94b27c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 23:35:07.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"521e676911cda8090e69f2c7aa544286548621ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 01:37:19.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcb636b20cbb45bd8ae53419f2b787a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbb0aacc166c4bdc8f3f164c536fe465 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd5681b2ff754610862a1b407d2931e3 : SUCCESS in 8m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19dd4ff9dbe04dcaa1c6a90b93e512be : SUCCESS in 11m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a616408ddff4aec975a356973bdaf2e : SUCCESS in 12m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/ef0b1b4ca5244dba92825b5a615cd9df : SUCCESS in 1h 02m 01s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/95bde9deb0f349a589a124b9c6102f1b : SUCCESS in 58m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2a83d903c6584ed19220687de78adff4 : SUCCESS in 1h 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a048ce997974f948058e56e62840d0f : SUCCESS in 24m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6086a653e4384737839261b1286f6b67 : SUCCESS in 7m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bfa13627cce144958a85374067e41c2c : SUCCESS in 1h 01m 26s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0a0a123f036e435b9c76ae8266e58c81 : SUCCESS in 1h 33m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa2106abf2d642e281029ed6941e5ebf : SUCCESS in 45m 09s","accounts_in_message":[],"_revision_number":12},{"id":"36297a9677d626b0ebcc7772d83dadb7922001f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 01:37:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"874df5b192dce3d619d8a78f84500e5dba2fed00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 01:39:11.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c1ef357c20d4db69f29b33316bcf072\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/335306e3b56f440d94135dc35ea25e04 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bec79834c2c848c5844277faf75fd9cb : SUCCESS in 36s","accounts_in_message":[],"_revision_number":12},{"id":"6fdca83427a9e54def7afdf70f9f164a1c02972b","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-06 06:00:08.000000000","message":"Patch Set 12:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/68/819668/12/fc : SUCCESS in 0h 52m 31s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/68/819668/12/iscsi : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"6a2716de161a1143b56c92bf1cc672c34573233a","revisions":{"8c7e5469e2e2655c2d898e714559049482d6cc6f":{"kind":"REWORK","_number":1,"created":"2021-11-29 18:19:05.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/68/819668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819668/1"}}},"commit":{"parents":[{"commit":"e5f5cc20641ea750818e7d3b22f91b67ea33d845","subject":"Remove consistencygroups quota entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e5f5cc20641ea750818e7d3b22f91b67ea33d845"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-10 11:42:44.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-29 18:17:26.000000000","tz":60},"subject":"Quota: Add backup related default limits","message":"Quota: Add backup related default limits\n\nLogs show deprecation warning in the logs when creating backups.\n\nWarnings are related to the \"backups\" and \"backup_gigabytes\" quota\nresources:\n\n  Deprecated: Default quota for resource: backups is set by the default\n  quota flag: quota_backups, it is now deprecated. Please use the\n  default quota class for default quota.\n\n  Deprecated: Default quota for resource: backup_gigabytes is set by the\n  default quota flag: quota_backup_gigabytes, it is now deprecated.\n  Please use the default quota class for default quota.\n\nThis warning is shown because the ``quota_classes`` table doesn\u0027t have\nentries for these 2 resources for the \"default\" quota class.\n\nThis patch adds a database migration to create the ``backups`` and\n``backup_gigabytes`` entries in the ``quota_classes`` table like we have\nfor the other standard resources, such as ``volumes``, ``gigabytes``,\netc.\n\nCloses-Bug: #1952420\nChange-Id: I2a83e9b23f40a8ef4a734b456e4b7afb1ad65f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c7e5469e2e2655c2d898e714559049482d6cc6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8c7e5469e2e2655c2d898e714559049482d6cc6f"}]},"branch":"refs/heads/master"},"3aa2a30ef7be0d11e53dbad2e699e327292a88f9":{"kind":"REWORK","_number":2,"created":"2021-11-30 17:25:30.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/68/819668/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819668/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819668/2"}}},"commit":{"parents":[{"commit":"a37c4c3fae4997a39b571bd2e659f423ea915bd8","subject":"Remove consistencygroups quota entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a37c4c3fae4997a39b571bd2e659f423ea915bd8"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-10 11:42:44.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-30 16:29:10.000000000","tz":60},"subject":"Quota: Add backup related default limits","message":"Quota: Add backup related default limits\n\nLogs show deprecation warning in the logs when creating backups.\n\nWarnings are related to the \"backups\" and \"backup_gigabytes\" quota\nresources:\n\n  Deprecated: Default quota for resource: backups is set by the default\n  quota flag: quota_backups, it is now deprecated. Please use the\n  default quota class for default quota.\n\n  Deprecated: Default quota for resource: backup_gigabytes is set by the\n  default quota flag: quota_backup_gigabytes, it is now deprecated.\n  Please use the default quota class for default quota.\n\nThis warning is shown because the ``quota_classes`` table doesn\u0027t have\nentries for these 2 resources for the \"default\" quota class.\n\nThis patch adds a database migration to create the ``backups`` and\n``backup_gigabytes`` entries in the ``quota_classes`` table like we have\nfor the other standard resources, such as ``volumes``, ``gigabytes``,\netc.\n\nCloses-Bug: #1952420\nChange-Id: I2a83e9b23f40a8ef4a734b456e4b7afb1ad65f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3aa2a30ef7be0d11e53dbad2e699e327292a88f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3aa2a30ef7be0d11e53dbad2e699e327292a88f9"}]},"branch":"refs/heads/master"},"0ee3555e994d544c65704d356771319066156192":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-02-14 14:25:16.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/68/819668/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819668/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819668/3"}}},"commit":{"parents":[{"commit":"968abe0c173e2498d6e6babeb9a2e5815d48ff1a","subject":"Remove consistencygroups quota entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/968abe0c173e2498d6e6babeb9a2e5815d48ff1a"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-10 11:42:44.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-02-14 14:24:37.000000000","tz":60},"subject":"Quota: Add backup related default limits","message":"Quota: Add backup related default limits\n\nLogs show deprecation warning in the logs when creating backups.\n\nWarnings are related to the \"backups\" and \"backup_gigabytes\" quota\nresources:\n\n  Deprecated: Default quota for resource: backups is set by the default\n  quota flag: quota_backups, it is now deprecated. Please use the\n  default quota class for default quota.\n\n  Deprecated: Default quota for resource: backup_gigabytes is set by the\n  default quota flag: quota_backup_gigabytes, it is now deprecated.\n  Please use the default quota class for default quota.\n\nThis warning is shown because the ``quota_classes`` table doesn\u0027t have\nentries for these 2 resources for the \"default\" quota class.\n\nThis patch adds a database migration to create the ``backups`` and\n``backup_gigabytes`` entries in the ``quota_classes`` table like we have\nfor the other standard resources, such as ``volumes``, ``gigabytes``,\netc.\n\nCloses-Bug: #1952420\nChange-Id: I2a83e9b23f40a8ef4a734b456e4b7afb1ad65f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0ee3555e994d544c65704d356771319066156192"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0ee3555e994d544c65704d356771319066156192"}]},"branch":"refs/heads/master"},"bcdcf8cfb77e1ef238bc6e3e6eb05f74307b9b3d":{"kind":"REWORK","_number":4,"created":"2022-05-26 11:40:00.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/68/819668/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819668/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819668/4"}}},"commit":{"parents":[{"commit":"08e7a6d53e2c0bb84c977d9f1602d1c4b689b151","subject":"Remove consistencygroups quota entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08e7a6d53e2c0bb84c977d9f1602d1c4b689b151"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-10 11:42:44.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-05-26 11:00:51.000000000","tz":120},"subject":"Quota: Add backup related default limits","message":"Quota: Add backup related default limits\n\nLogs show deprecation warning in the logs when creating backups.\n\nWarnings are related to the \"backups\" and \"backup_gigabytes\" quota\nresources:\n\n  Deprecated: Default quota for resource: backups is set by the default\n  quota flag: quota_backups, it is now deprecated. Please use the\n  default quota class for default quota.\n\n  Deprecated: Default quota for resource: backup_gigabytes is set by the\n  default quota flag: quota_backup_gigabytes, it is now deprecated.\n  Please use the default quota class for default quota.\n\nThis warning is shown because the ``quota_classes`` table doesn\u0027t have\nentries for these 2 resources for the \"default\" quota class.\n\nThis patch adds a database migration to create the ``backups`` and\n``backup_gigabytes`` entries in the ``quota_classes`` table like we have\nfor the other standard resources, such as ``volumes``, ``gigabytes``,\netc.\n\nCloses-Bug: #1952420\nChange-Id: I2a83e9b23f40a8ef4a734b456e4b7afb1ad65f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bcdcf8cfb77e1ef238bc6e3e6eb05f74307b9b3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bcdcf8cfb77e1ef238bc6e3e6eb05f74307b9b3d"}]},"branch":"refs/heads/master"},"d156b4661fc908812f5de1758552d50350569022":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-08-29 15:07:25.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/68/819668/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819668/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819668/5"}}},"commit":{"parents":[{"commit":"37ad5d8728def3ca728d3404d027e67cb72abf67","subject":"Remove consistencygroups quota entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/37ad5d8728def3ca728d3404d027e67cb72abf67"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-10 11:42:44.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-29 13:53:28.000000000","tz":120},"subject":"Quota: Add backup related default limits","message":"Quota: Add backup related default limits\n\nLogs show deprecation warning in the logs when creating backups.\n\nWarnings are related to the \"backups\" and \"backup_gigabytes\" quota\nresources:\n\n  Deprecated: Default quota for resource: backups is set by the default\n  quota flag: quota_backups, it is now deprecated. Please use the\n  default quota class for default quota.\n\n  Deprecated: Default quota for resource: backup_gigabytes is set by the\n  default quota flag: quota_backup_gigabytes, it is now deprecated.\n  Please use the default quota class for default quota.\n\nThis warning is shown because the ``quota_classes`` table doesn\u0027t have\nentries for these 2 resources for the \"default\" quota class.\n\nThis patch adds a database migration to create the ``backups`` and\n``backup_gigabytes`` entries in the ``quota_classes`` table like we have\nfor the other standard resources, such as ``volumes``, ``gigabytes``,\netc.\n\nCloses-Bug: #1952420\nChange-Id: I2a83e9b23f40a8ef4a734b456e4b7afb1ad65f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d156b4661fc908812f5de1758552d50350569022"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d156b4661fc908812f5de1758552d50350569022"}]},"branch":"refs/heads/master"},"a3bd5ac5dcfbf96ecdd7768bd641238d30d11237":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-08-29 15:12:07.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/68/819668/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819668/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819668/6"}}},"commit":{"parents":[{"commit":"fcfc46db070db2dc348b992816a556e828619f92","subject":"Remove consistencygroups quota entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fcfc46db070db2dc348b992816a556e828619f92"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-10 11:42:44.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-29 15:11:50.000000000","tz":120},"subject":"Quota: Add backup related default limits","message":"Quota: Add backup related default limits\n\nLogs show deprecation warning in the logs when creating backups.\n\nWarnings are related to the \"backups\" and \"backup_gigabytes\" quota\nresources:\n\n  Deprecated: Default quota for resource: backups is set by the default\n  quota flag: quota_backups, it is now deprecated. Please use the\n  default quota class for default quota.\n\n  Deprecated: Default quota for resource: backup_gigabytes is set by the\n  default quota flag: quota_backup_gigabytes, it is now deprecated.\n  Please use the default quota class for default quota.\n\nThis warning is shown because the ``quota_classes`` table doesn\u0027t have\nentries for these 2 resources for the \"default\" quota class.\n\nThis patch adds a database migration to create the ``backups`` and\n``backup_gigabytes`` entries in the ``quota_classes`` table like we have\nfor the other standard resources, such as ``volumes``, ``gigabytes``,\netc.\n\nCloses-Bug: #1952420\nChange-Id: I2a83e9b23f40a8ef4a734b456e4b7afb1ad65f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a3bd5ac5dcfbf96ecdd7768bd641238d30d11237"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a3bd5ac5dcfbf96ecdd7768bd641238d30d11237"}]},"branch":"refs/heads/master"},"231de3ac5ebb4aea7dbe67d33009653bf613be26":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-03-14 18:08:06.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/68/819668/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819668/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819668/7"}}},"commit":{"parents":[{"commit":"6c64d0b2f1196e8f5461f59e833ad53f5cb4dde9","subject":"Remove consistencygroups quota entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c64d0b2f1196e8f5461f59e833ad53f5cb4dde9"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-10 11:42:44.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2023-03-14 11:51:47.000000000","tz":60},"subject":"Quota: Add backup related default limits","message":"Quota: Add backup related default limits\n\nLogs show deprecation warning in the logs when creating backups.\n\nWarnings are related to the \"backups\" and \"backup_gigabytes\" quota\nresources:\n\n  Deprecated: Default quota for resource: backups is set by the default\n  quota flag: quota_backups, it is now deprecated. Please use the\n  default quota class for default quota.\n\n  Deprecated: Default quota for resource: backup_gigabytes is set by the\n  default quota flag: quota_backup_gigabytes, it is now deprecated.\n  Please use the default quota class for default quota.\n\nThis warning is shown because the ``quota_classes`` table doesn\u0027t have\nentries for these 2 resources for the \"default\" quota class.\n\nThis patch adds a database migration to create the ``backups`` and\n``backup_gigabytes`` entries in the ``quota_classes`` table like we have\nfor the other standard resources, such as ``volumes``, ``gigabytes``,\netc.\n\nCloses-Bug: #1952420\nChange-Id: I2a83e9b23f40a8ef4a734b456e4b7afb1ad65f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/231de3ac5ebb4aea7dbe67d33009653bf613be26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/231de3ac5ebb4aea7dbe67d33009653bf613be26"}]},"branch":"refs/heads/master"},"4cd068e90f6e4506358a632c4b6eb79bf7bb26d8":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-10-20 08:16:15.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/68/819668/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819668/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819668/8"}}},"commit":{"parents":[{"commit":"4aec2d14765520f0bb1c04b6622b0a93da0027fb","subject":"Remove consistencygroups quota entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4aec2d14765520f0bb1c04b6622b0a93da0027fb"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-10 11:42:44.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2023-10-19 18:24:11.000000000","tz":120},"subject":"Quota: Add backup related default limits","message":"Quota: Add backup related default limits\n\nLogs show deprecation warning in the logs when creating backups.\n\nWarnings are related to the \"backups\" and \"backup_gigabytes\" quota\nresources:\n\n  Deprecated: Default quota for resource: backups is set by the default\n  quota flag: quota_backups, it is now deprecated. Please use the\n  default quota class for default quota.\n\n  Deprecated: Default quota for resource: backup_gigabytes is set by the\n  default quota flag: quota_backup_gigabytes, it is now deprecated.\n  Please use the default quota class for default quota.\n\nThis warning is shown because the ``quota_classes`` table doesn\u0027t have\nentries for these 2 resources for the \"default\" quota class.\n\nThis patch adds a database migration to create the ``backups`` and\n``backup_gigabytes`` entries in the ``quota_classes`` table like we have\nfor the other standard resources, such as ``volumes``, ``gigabytes``,\netc.\n\nCloses-Bug: #1952420\nChange-Id: I2a83e9b23f40a8ef4a734b456e4b7afb1ad65f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4cd068e90f6e4506358a632c4b6eb79bf7bb26d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4cd068e90f6e4506358a632c4b6eb79bf7bb26d8"}]},"branch":"refs/heads/master"},"0c4edf415f2d9f8013066235b1071211bb1eeb90":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-11-13 16:46:14.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/68/819668/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819668/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819668/9"}}},"commit":{"parents":[{"commit":"96aa3221160c21f8493e33c37bf74446351a2feb","subject":"Remove consistencygroups quota entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96aa3221160c21f8493e33c37bf74446351a2feb"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-10 11:42:44.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2023-11-13 15:01:14.000000000","tz":60},"subject":"Quota: Add backup related default limits","message":"Quota: Add backup related default limits\n\nLogs show deprecation warning in the logs when creating backups.\n\nWarnings are related to the \"backups\" and \"backup_gigabytes\" quota\nresources:\n\n  Deprecated: Default quota for resource: backups is set by the default\n  quota flag: quota_backups, it is now deprecated. Please use the\n  default quota class for default quota.\n\n  Deprecated: Default quota for resource: backup_gigabytes is set by the\n  default quota flag: quota_backup_gigabytes, it is now deprecated.\n  Please use the default quota class for default quota.\n\nThis warning is shown because the ``quota_classes`` table doesn\u0027t have\nentries for these 2 resources for the \"default\" quota class.\n\nThis patch adds a database migration to create the ``backups`` and\n``backup_gigabytes`` entries in the ``quota_classes`` table like we have\nfor the other standard resources, such as ``volumes``, ``gigabytes``,\netc.\n\nCloses-Bug: #1952420\nChange-Id: I2a83e9b23f40a8ef4a734b456e4b7afb1ad65f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0c4edf415f2d9f8013066235b1071211bb1eeb90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0c4edf415f2d9f8013066235b1071211bb1eeb90"}]},"branch":"refs/heads/master"},"2b3020fd670dc41dbe8714292d2929b28d9287d9":{"kind":"REWORK","_number":10,"created":"2024-01-12 13:13:24.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/68/819668/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819668/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819668/10"}}},"commit":{"parents":[{"commit":"857b359cfa865e0d06afca281680f2632df1a26a","subject":"Remove consistencygroups quota entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/857b359cfa865e0d06afca281680f2632df1a26a"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-10 11:42:44.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2024-01-12 11:18:23.000000000","tz":60},"subject":"Quota: Add backup related default limits","message":"Quota: Add backup related default limits\n\nLogs show deprecation warning in the logs when creating backups.\n\nWarnings are related to the \"backups\" and \"backup_gigabytes\" quota\nresources:\n\n  Deprecated: Default quota for resource: backups is set by the default\n  quota flag: quota_backups, it is now deprecated. Please use the\n  default quota class for default quota.\n\n  Deprecated: Default quota for resource: backup_gigabytes is set by the\n  default quota flag: quota_backup_gigabytes, it is now deprecated.\n  Please use the default quota class for default quota.\n\nThis warning is shown because the ``quota_classes`` table doesn\u0027t have\nentries for these 2 resources for the \"default\" quota class.\n\nThis patch adds a database migration to create the ``backups`` and\n``backup_gigabytes`` entries in the ``quota_classes`` table like we have\nfor the other standard resources, such as ``volumes``, ``gigabytes``,\netc.\n\nCloses-Bug: #1952420\nChange-Id: I2a83e9b23f40a8ef4a734b456e4b7afb1ad65f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b3020fd670dc41dbe8714292d2929b28d9287d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b3020fd670dc41dbe8714292d2929b28d9287d9"}]},"branch":"refs/heads/master"},"d46f3673cbe3c070088024f8462a51b690e41009":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-01-12 13:26:05.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/68/819668/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819668/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819668/11"}}},"commit":{"parents":[{"commit":"3dc27f8b6e507a214e4aa2a77c87694a6938823d","subject":"Remove consistencygroups quota entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3dc27f8b6e507a214e4aa2a77c87694a6938823d"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-10 11:42:44.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2024-01-12 13:25:44.000000000","tz":60},"subject":"Quota: Add backup related default limits","message":"Quota: Add backup related default limits\n\nLogs show deprecation warning in the logs when creating backups.\n\nWarnings are related to the \"backups\" and \"backup_gigabytes\" quota\nresources:\n\n  Deprecated: Default quota for resource: backups is set by the default\n  quota flag: quota_backups, it is now deprecated. Please use the\n  default quota class for default quota.\n\n  Deprecated: Default quota for resource: backup_gigabytes is set by the\n  default quota flag: quota_backup_gigabytes, it is now deprecated.\n  Please use the default quota class for default quota.\n\nThis warning is shown because the ``quota_classes`` table doesn\u0027t have\nentries for these 2 resources for the \"default\" quota class.\n\nThis patch adds a database migration to create the ``backups`` and\n``backup_gigabytes`` entries in the ``quota_classes`` table like we have\nfor the other standard resources, such as ``volumes``, ``gigabytes``,\netc.\n\nCloses-Bug: #1952420\nChange-Id: I2a83e9b23f40a8ef4a734b456e4b7afb1ad65f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d46f3673cbe3c070088024f8462a51b690e41009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d46f3673cbe3c070088024f8462a51b690e41009"}]},"branch":"refs/heads/master"},"6a2716de161a1143b56c92bf1cc672c34573233a":{"kind":"REWORK","_number":12,"created":"2024-01-17 18:10:12.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/68/819668/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/819668/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/819668/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/819668/12"}}},"commit":{"parents":[{"commit":"2b88148c6c684d9ac6488ce836978fb16fc9a429","subject":"Remove consistencygroups quota entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b88148c6c684d9ac6488ce836978fb16fc9a429"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-10 11:42:44.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2024-01-17 18:07:42.000000000","tz":60},"subject":"Quota: Add backup related default limits","message":"Quota: Add backup related default limits\n\nLogs show deprecation warning in the logs when creating backups.\n\nWarnings are related to the \"backups\" and \"backup_gigabytes\" quota\nresources:\n\n  Deprecated: Default quota for resource: backups is set by the default\n  quota flag: quota_backups, it is now deprecated. Please use the\n  default quota class for default quota.\n\n  Deprecated: Default quota for resource: backup_gigabytes is set by the\n  default quota flag: quota_backup_gigabytes, it is now deprecated.\n  Please use the default quota class for default quota.\n\nThis warning is shown because the ``quota_classes`` table doesn\u0027t have\nentries for these 2 resources for the \"default\" quota class.\n\nThis patch adds a database migration to create the ``backups`` and\n``backup_gigabytes`` entries in the ``quota_classes`` table like we have\nfor the other standard resources, such as ``volumes``, ``gigabytes``,\netc.\n\nCloses-Bug: #1952420\nChange-Id: I2a83e9b23f40a8ef4a734b456e4b7afb1ad65f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6a2716de161a1143b56c92bf1cc672c34573233a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6a2716de161a1143b56c92bf1cc672c34573233a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
