)]}'
{"id":"openstack%2Fcinder~819691","triplet_id":"openstack%2Fcinder~master~Iedc9b8939ffcda2ca7fad0357aac97ae9a2df588","project":"openstack/cinder","branch":"master","topic":"die-quota-die","attention_set":{},"removed_from_attention_set":{"33431":{"account":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"last_update":"2024-01-29 22:53:34.000000000","reason":"Change was abandoned"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2024-01-29 22:53:34.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iedc9b8939ffcda2ca7fad0357aac97ae9a2df588","subject":"WIP: Quota: New quota system","status":"ABANDONED","created":"2021-11-29 18:19:05.000000000","updated":"2024-01-29 22:53:34.000000000","total_comment_count":15,"unresolved_comment_count":12,"has_review_started":true,"meta_rev_id":"fa441065f949b1a6ab348bf9af7d56f8276d4b16","_number":819691,"virtual_id_number":819691,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-09 15:39:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-01-29 22:52:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-29 18:24:06.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-11-29 18:26:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-29 18:36:18.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-29 18:42:12.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-29 18:49:41.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-29 22:17:26.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-11-30 01:33:52.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-30 05:09:44.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-11-30 10:26:35.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-30 11:34:37.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-30 13:36:55.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-11-30 22:48:16.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-12-02 18:51:50.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2021-12-05 17:11:34.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-06 07:44:04.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-09 15:02:54.000000000","updated_by":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"reviewer":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"state":"REVIEWER"},{"updated":"2022-11-24 08:27:50.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2024-01-29 22:52:59.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"c4c82c486e99e550f3bb112ac57bf9bc51f71a56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-11-29 18:19:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a15f6b9ab9add6c77c83fb6f3f4713b057177912","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-11-29 18:24:06.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@master\", line 63, column 32","accounts_in_message":[],"_revision_number":1},{"id":"02b65c9ac218d417cb3a93522913adb57c55ff9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 18:26:04.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@master\", line 63, column 32","accounts_in_message":[],"_revision_number":1},{"id":"01c7f82d9e99e069a485ab7609530d9ab6f2a182","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 18:28:24.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@master\", line 63, column 32","accounts_in_message":[],"_revision_number":1},{"id":"7444f203a2a1d1fb420b782f5589aff92db6ee9a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-29 18:36:18.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@master\", line 63, column 32","accounts_in_message":[],"_revision_number":1},{"id":"0b4e0c00dd41a9b3a95141701f44198498a6aad6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-29 18:37:16.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@master\", line 63, column 32","accounts_in_message":[],"_revision_number":1},{"id":"1d55eb7d012371f2d7fafe325d60d16e027467e5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-11-29 18:42:12.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@master\", line 63, column 32","accounts_in_message":[],"_revision_number":1},{"id":"a8f9378821a2506b8cfff12d6da89cfc0bb3a1b2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-29 18:49:41.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  found undefined alias \u0027gate-irrelevant-files\u0027\n  in \"openstack/cinder/.zuul.yaml@master\", line 63, column 32","accounts_in_message":[],"_revision_number":1},{"id":"9d8df9b2100b5bc9d59399bb83b09da495782f25","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-29 22:17:26.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819691/1/iscsi/ : FAILURE in 1h 41m 35s\n- cinder-smb http://cloudbase-ci.com/cinder/819691/1/smb/ : SUCCESS in 1h 03m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e074a17519ffc3a1bfb71bafcb44a41833895fe5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-30 01:33:52.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-91-819691-1/ns5_nfs : FAILURE in 1h 59m 2s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-91-819691-1/ns5_iscsi : FAILURE in 1h 57m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"6d33cfdc968fa26dae53a3fb170a8a543740212a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-11-30 05:09:44.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/91/819691/1/check/DellEMC_Unity_FC/0fe3879/DellEMC_Unity_FC/209 : FAILURE in 1h 05m 54s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/91/819691/1/check/DellEMC_Unity_iSCSI/fef0b46/DellEMC_Unity_iSCSI/211 : FAILURE in 2m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"e73e469c1d68ec597f961e39da6b763ed741e38b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-30 10:26:35.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/91/819691/1/check/DellEMC_XtremIO_FC/cbb643b/DellEMC_XtremIO_FC/248 : SUCCESS in 1h 07m 17s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/91/819691/1/check/DellEMC_XtremIO_iSCSI/00177b5/DellEMC_XtremIO_iSCSI/265 : SUCCESS in 1h 47m 27s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"ea22ced54cf946ee67f84d8d2830eba7ceaac697","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-30 11:34:37.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"0ce64227aee1193e233133230f42418da3221f7b","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-11-30 13:36:55.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/91/819691/1/check/dsvm-tempest-TOYOU-ACS5000-driver/ac963ba : FAILURE in 2h 03m 41s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/91/819691/1/check/fc-test/b69fae7 : SUCCESS in 1h 56m 55s","accounts_in_message":[],"_revision_number":1},{"id":"01f4ccc2bc827d0db4c10dedbbaacf4a314b6a8d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-30 15:24:09.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"1d0173b00dfde12e4480bc55b9eb72ee7f34ceff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-11-30 17:25:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bf1c7377735b72ccba4fa8a1c51972414f5b4474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 17:34:30.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027experimental\u0027][\u0027jobs\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python3-yoga-jobs\n        - openstack-python3-yoga-jobs-arm64\n        - publish-openstack-docs-pti\n        - periodic-stable-jobs\n        - check-requirements\n        - integrated-gate-storage\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          # - cinder-code-coverage:\n          #     voting: false\n          # - cinder-mypy:\n          #     voting: false\n          # - cinder-tox-bandit-baseline:\n          #     voting: false\n          # - openstack-tox-functional-py36:\n          #     irrelevant-files: \u0026functional-irrelevant-files\n          #       - ^.*\\.rst$\n          #       - ^cinder/locale/.*$\n          #       - ^cinder/tests/hacking/.*$\n          #       - ^cinder/tests/unit.*$\n          #       - ^contrib/block-box.*$\n          #       - ^doc/.*$\n          #       - ^releasenotes/.*$\n          # - openstack-tox-functional-py38:\n          #     irrelevant-files: *functional-irrelevant-files\n          # - cinder-rally-task:\n          #     voting: false\n          # - openstack-tox-pylint:\n          #     voting: false\n          #     timeout: 5400\n          #     irrelevant-files:\n          #       - ^.*\\.rst$\n          #       - ^api-ref/.*$\n          #       - ^cinder/locale/.*$\n          #       - ^cinder/tests/hacking/.*$\n          #       - ^cinder/tests/unit.*$\n          #       - ^contrib/block-box.*$\n          #       - ^doc/.*$\n          #       - ^releasenotes/.*$\n          # - cinder-plugin-ceph-tempest:\n          #     irrelevant-files: \u0026gate-irrelevant-files\n          #       - ^(test-|)requirements.txt$\n          #       - ^.*\\.rst$\n          #       - ^api-ref/.*$\n          #       - ^cinder/cmd/status\\.py$\n          #       - ^cinder/locale/.*$\n          #       - ^cinder/tests/functional.*$\n          #       - ^cinder/tests/hacking/.*$\n          #       - ^cinder/tests/unit.*$\n          #       - ^contrib/block-box.*$\n          #       - ^doc/.*$\n          #       - ^releasenotes/.*$\n          #       - ^setup.cfg$\n          #       - ^tools/.*$\n          #       - ^tox.ini$\n          # - cinder-plugin-ceph-tempest-mn-aa:\n          #     voting: false\n          #     irrelevant-files: *gate-irrelevant-files\n          - cinder-tempest-plugin-lvm-lio-barbican:\n               irrelevant-files: \u0026gate-irrelevant-files\n                 - ^(test-|)requirements.txt$\n                 - ^.*\\.rst$\n                 - ^api-ref/.*$\n                 - ^cinder/cmd/status\\.py$\n                 - ^cinder/locale/.*$\n                 - ^cinder/tests/functional.*$\n                 - ^cinder/tests/hacking/.*$\n                 - ^cinder/tests/unit.*$\n                 - ^contrib/block-box.*$\n                 - ^doc/.*$\n                 - ^releasenotes/.*$\n                 - ^setup.cfg$\n                 - ^tools/.*$\n                 - ^tox.ini$\n          # - cinder-grenade-mn-sub-volbak:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - cinder-tempest-lvm-multibackend:\n          #     voting: false\n          #     irrelevant-files: *gate-irrelevant-files\n          # - cinder-for-glance-optimized:\n          #     voting: false\n          #     irrelevant-files: *gate-irrelevant-files\n          # - devstack-plugin-nfs-tempest-full:\n          #     voting: false\n          #     irrelevant-files: *gate-irrelevant-files\n          # - tempest-slow-py3:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - tempest-integrated-storage:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - grenade:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - tempest-ipv6-only:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - openstacksdk-functional-devstack:\n          #     irrelevant-files: *gate-irrelevant-files\n      gate:\n        jobs:\n          - cinder-grenade-mn-sub-volbak:\n              irrelevant-files: *gate-irrelevant-files\n          - cinder-plugin-ceph-tempest:\n              irrelevant-files: *gate-irrelevant-files\n          - tempest-integrated-storage:\n              irrelevant-files: *gate-irrelevant-files\n          - grenade:\n              irrelevant-files: *gate-irrelevant-files\n          - tempest-ipv6-only:\n              irrelevant-files: *gate-irrelevant-files\n          - openstacksdk-functional-devstack:\n              irrelevant-files: *gate-irrelevant-files\n      experimental:\n        jobs:\n          # - cinder-multibackend-matrix-migration:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - cinder-grenade-mn-sub-volschbak:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - cinder-grenade-mn-sub-bak:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - devstack-plugin-ceph-tempest-py3:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - tempest-pg-full:\n          #     irrelevant-files: *gate-irrelevant-files\n\n  in \"openstack/cinder/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":2},{"id":"0c6f13773410a6ecae3fc6cd405804fb0fac8018","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 17:36:16.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027experimental\u0027][\u0027jobs\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python3-yoga-jobs\n        - openstack-python3-yoga-jobs-arm64\n        - publish-openstack-docs-pti\n        - periodic-stable-jobs\n        - check-requirements\n        - integrated-gate-storage\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          # - cinder-code-coverage:\n          #     voting: false\n          # - cinder-mypy:\n          #     voting: false\n          # - cinder-tox-bandit-baseline:\n          #     voting: false\n          # - openstack-tox-functional-py36:\n          #     irrelevant-files: \u0026functional-irrelevant-files\n          #       - ^.*\\.rst$\n          #       - ^cinder/locale/.*$\n          #       - ^cinder/tests/hacking/.*$\n          #       - ^cinder/tests/unit.*$\n          #       - ^contrib/block-box.*$\n          #       - ^doc/.*$\n          #       - ^releasenotes/.*$\n          # - openstack-tox-functional-py38:\n          #     irrelevant-files: *functional-irrelevant-files\n          # - cinder-rally-task:\n          #     voting: false\n          # - openstack-tox-pylint:\n          #     voting: false\n          #     timeout: 5400\n          #     irrelevant-files:\n          #       - ^.*\\.rst$\n          #       - ^api-ref/.*$\n          #       - ^cinder/locale/.*$\n          #       - ^cinder/tests/hacking/.*$\n          #       - ^cinder/tests/unit.*$\n          #       - ^contrib/block-box.*$\n          #       - ^doc/.*$\n          #       - ^releasenotes/.*$\n          # - cinder-plugin-ceph-tempest:\n          #     irrelevant-files: \u0026gate-irrelevant-files\n          #       - ^(test-|)requirements.txt$\n          #       - ^.*\\.rst$\n          #       - ^api-ref/.*$\n          #       - ^cinder/cmd/status\\.py$\n          #       - ^cinder/locale/.*$\n          #       - ^cinder/tests/functional.*$\n          #       - ^cinder/tests/hacking/.*$\n          #       - ^cinder/tests/unit.*$\n          #       - ^contrib/block-box.*$\n          #       - ^doc/.*$\n          #       - ^releasenotes/.*$\n          #       - ^setup.cfg$\n          #       - ^tools/.*$\n          #       - ^tox.ini$\n          # - cinder-plugin-ceph-tempest-mn-aa:\n          #     voting: false\n          #     irrelevant-files: *gate-irrelevant-files\n          - cinder-tempest-plugin-lvm-lio-barbican:\n               irrelevant-files: \u0026gate-irrelevant-files\n                 - ^(test-|)requirements.txt$\n                 - ^.*\\.rst$\n                 - ^api-ref/.*$\n                 - ^cinder/cmd/status\\.py$\n                 - ^cinder/locale/.*$\n                 - ^cinder/tests/functional.*$\n                 - ^cinder/tests/hacking/.*$\n                 - ^cinder/tests/unit.*$\n                 - ^contrib/block-box.*$\n                 - ^doc/.*$\n                 - ^releasenotes/.*$\n                 - ^setup.cfg$\n                 - ^tools/.*$\n                 - ^tox.ini$\n          # - cinder-grenade-mn-sub-volbak:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - cinder-tempest-lvm-multibackend:\n          #     voting: false\n          #     irrelevant-files: *gate-irrelevant-files\n          # - cinder-for-glance-optimized:\n          #     voting: false\n          #     irrelevant-files: *gate-irrelevant-files\n          # - devstack-plugin-nfs-tempest-full:\n          #     voting: false\n          #     irrelevant-files: *gate-irrelevant-files\n          # - tempest-slow-py3:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - tempest-integrated-storage:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - grenade:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - tempest-ipv6-only:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - openstacksdk-functional-devstack:\n          #     irrelevant-files: *gate-irrelevant-files\n      gate:\n        jobs:\n          - cinder-grenade-mn-sub-volbak:\n              irrelevant-files: *gate-irrelevant-files\n          - cinder-plugin-ceph-tempest:\n              irrelevant-files: *gate-irrelevant-files\n          - tempest-integrated-storage:\n              irrelevant-files: *gate-irrelevant-files\n          - grenade:\n              irrelevant-files: *gate-irrelevant-files\n          - tempest-ipv6-only:\n              irrelevant-files: *gate-irrelevant-files\n          - openstacksdk-functional-devstack:\n              irrelevant-files: *gate-irrelevant-files\n      experimental:\n        jobs:\n          # - cinder-multibackend-matrix-migration:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - cinder-grenade-mn-sub-volschbak:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - cinder-grenade-mn-sub-bak:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - devstack-plugin-ceph-tempest-py3:\n          #     irrelevant-files: *gate-irrelevant-files\n          # - tempest-pg-full:\n          #     irrelevant-files: *gate-irrelevant-files\n\n  in \"openstack/cinder/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":2},{"id":"97749fbfb4f04a821829682869a5c1ceef32083a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-11-30 18:08:05.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819691/2/iscsi/ : FAILURE in 13m 54s\n- cinder-smb http://cloudbase-ci.com/cinder/819691/2/smb/ : FAILURE in 13m 49s","accounts_in_message":[],"_revision_number":2},{"id":"9f4c0e7341d35e89208c4894a0bd8e754848711f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-11-30 20:11:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k finger://openstackcifc-virtual-machine/7c4b6b04f31442a48d3e2fa5b10b19e7 : RETRY_LIMIT in 10s","accounts_in_message":[],"_revision_number":2},{"id":"bbbf96a16a39ccc5c3727c710e7498f95ad4316e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-30 21:48:48.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/91/819691/2/check/devstack-plugin-ceph-dsvm-focal-py3/8b4b804/ : RETRY_LIMIT in 23m 17s","accounts_in_message":[],"_revision_number":2},{"id":"8786f1470b3cec1eb439256355c4ffffdc9dccc1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-30 22:48:16.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/91/819691/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/607f69a/ : FAILURE in 34m 34s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/91/819691/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/91a29c3/ : FAILURE in 43m 44s","accounts_in_message":[],"_revision_number":2},{"id":"c66de22d8d3ad1826f30c76e99d40536af09d4d9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-01 00:23:13.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/2/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"d6d8b6414ca1bd79fb4ceafe403f170d9b25e0c8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-01 02:07:03.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-91-819691-2/ns5_nfs : FAILURE in 1h 58m 8s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-91-819691-2/ns5_iscsi : FAILURE in 1h 53m 37s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"f19871eba8b5faebc3f2439f9c456baf470e4bc6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-12-01 02:33:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/91/819691/2/check/dsvm-tempest-cisco-zonemanager-job/f409e6f/ : FAILURE in 1h 04m 00s","accounts_in_message":[],"_revision_number":2},{"id":"e4644ecbebc233f60b6daf908b88eafde40c0a54","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-12-01 03:02:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/91/819691/2/check/iscsi-driver-nimble/cdecb85/cdecb85.html : FAILURE in 58m 12s","accounts_in_message":[],"_revision_number":2},{"id":"5a5b608d9f29b7bcda87f87f47046c4bca7efbdb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-01 09:18:15.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/91/819691/2/check/DellEMC_Unity_FC/aca0a06/DellEMC_Unity_FC/229 : FAILURE in 1h 09m 33s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/91/819691/2/check/DellEMC_Unity_iSCSI/1afcb43/DellEMC_Unity_iSCSI/231 : SUCCESS in 1h 23m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"0d73d618f3003eb04e186478a1b83f5b6cd1b0c3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-01 09:28:46.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"31b238ce5c32393903ba68ce130db5d1453e2ac3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-12-01 10:17:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a2801d03e16ebd2ec09a9a9a41669ff9b4114fb3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-12-01 10:19:01.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"371c6158f68ddcf26608c1ef8863d9933c4fa7d9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-01 10:33:04.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819691/3/iscsi/ : FAILURE in 13m 49s\n- cinder-smb http://cloudbase-ci.com/cinder/819691/3/smb/ : FAILURE in 13m 55s","accounts_in_message":[],"_revision_number":3},{"id":"e0cdcdcb8fedef088e08a5f3b750582f6f95eaf1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-01 10:34:52.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"a16c6ecb745a29762119252a861ed05acf07e69b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 10:40:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/71d37826e0e44349a3c7c0f5a9c8dee9 : FAILURE in 13m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f402db06cabc4de394750c9b6bfbe48b : FAILURE in 13m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6bb23b6b5c4e09086edf056beccc0bea8a728f17","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-01 10:58:32.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/91/819691/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8be9374/ : FAILURE in 38m 39s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/91/819691/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/c0ebcfe/ : FAILURE in 38m 24s","accounts_in_message":[],"_revision_number":3},{"id":"729e9b34e2b8f647295a0744a915b9b93def9b2f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-12-01 11:17:23.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/91/819691/3/check/iscsi-driver-alletra-6k/0b4cbbd/0b4cbbd.html : FAILURE in 53m 11s","accounts_in_message":[],"_revision_number":3},{"id":"b528bbea0c0ec70807357445e7fa7bfa4d863886","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2021-12-01 11:22:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/91/819691/3/check/iscsi-driver-nimble/e70bafc/e70bafc.html : FAILURE in 57m 35s","accounts_in_message":[],"_revision_number":3},{"id":"b00d79f03edfa9ec67963081d68aced77c277f5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 11:37:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e7a2ba6d2dd49e2bd5585795beceb0d : FAILURE in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86786d8df7ba4f09aae426d4699672c5 : FAILURE in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c58482e8f7545a58ae0fb027d880adb : FAILURE in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80a305c0846d41d581a5ebf4f27a65a9 : SUCCESS in 11m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/4b759d49a2b84a388562041885aa3a4a : FAILURE in 1h 01m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87f579b9fe9742c5982557e173731582 : FAILURE in 57m 17s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d9c79ee28508420599a2bcd501cc2ea4 : FAILURE in 1h 09m 11s","accounts_in_message":[],"_revision_number":3},{"id":"8d36273996656b92c1ba957765d1e9f7f245b0f6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-01 11:53:06.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-91-819691-3/ns5_nfs : FAILURE in 1h 34m 39s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-91-819691-3/ns5_iscsi : FAILURE in 1h 35m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"293230624e77ada7918bc3e740347022a5e0ad2d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-01 12:32:28.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/91/819691/3/check/tempest-dsvm-full-focal-py3/1742c05/ : FAILURE in 2h 05m 14s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/91/819691/3/check/devstack-plugin-ceph-dsvm-focal-py3/d5ee8ee/ : RETRY_LIMIT in 3m 03s","accounts_in_message":[],"_revision_number":3},{"id":"0839cd8d59a6cdd83a6602126c25fc09d678fc70","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-01 15:32:10.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/91/819691/3/check/DellEMC_Unity_FC/9aad91d/DellEMC_Unity_FC/234 : FAILURE in 42m 09s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/91/819691/3/check/DellEMC_Unity_iSCSI/2ca5a6b/DellEMC_Unity_iSCSI/236 : FAILURE in 47m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"51d5f78dd62bae33de2e1a37dc7e7b9cacb79cfe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-01 16:21:16.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"e972261f773aae8fe8f33efab60d9ee01031f5fc","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-01 20:19:41.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/91/819691/3/check/DellEMC_XtremIO_FC/ceb2576/DellEMC_XtremIO_FC/271 : FAILURE in 59m 56s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/91/819691/3/check/DellEMC_XtremIO_iSCSI/15340c9/DellEMC_XtremIO_iSCSI/286 : FAILURE in 1h 07m 28s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"943eb59fe0ef164486654dcb313a65c34fd1c7de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-01 22:24:20.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/3/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"273d4783b53e65d5fe72535384c1af0a3a2d5415","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-02 00:47:15.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b3bfeabf494046d982e9090169194b2d : FAILURE in 1h 00m 35s","accounts_in_message":[],"_revision_number":3},{"id":"4eebc48e551c841e411817d4e599a0fac3e57b51","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-12-02 17:55:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8b0dba63954a543916c8cb19e3cca653f9bbda17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-12-02 17:55:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b96a414e4d74b4f68f49a48c955b9247ef71e957","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-02 18:10:54.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819691/4/iscsi/ : FAILURE in 13m 47s\n- cinder-smb http://cloudbase-ci.com/cinder/819691/4/smb/ : FAILURE in 13m 42s","accounts_in_message":[],"_revision_number":4},{"id":"3dd31cb77d70f6121823187ec6b479d4282dd745","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 18:17:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/69fe91fdca904b83b2d423b02eae6223 : FAILURE in 8m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/65f1e0601a644c6d92cc2188024b0872 : FAILURE in 17m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"262bcfc9945b9225a0ccd0c23a23f7cf5a7a4762","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-02 18:51:50.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/91/819691/4/check/iscsi-driver-alletra-6k/2cad11d/2cad11d.html : FAILURE in 52m 05s","accounts_in_message":[],"_revision_number":4},{"id":"1121d8f893fb945a529fca150607278b435be244","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 19:00:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/890cb62c1b9f4af9bb5c9e158edf96fa : FAILURE in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/401cbc81dfd04016b268eb836f91ac73 : FAILURE in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b03834ae60db4f8b9b787792075957cc : FAILURE in 12m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ed25182856b4530bf92d350e84a9002 : SUCCESS in 11m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/4b864c5041b341fe97e4dd61feca37df : FAILURE in 45m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ef1f9082feb4af691949a3493c9c472 : FAILURE in 46m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e4b9066e01842d28ea3b1a32f40400f : FAILURE in 37m 06s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bab4ea510cdd490c81de895a7c5e0c71 : FAILURE in 58m 12s","accounts_in_message":[],"_revision_number":4},{"id":"2a2d4fa40d6152a3e944f134de3b9ea3cde74106","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-02 19:49:21.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/91/819691/2/check/dsvm-tempest-TOYOU-ACS5000-driver/86e9f35 : FAILURE in 2h 27m 13s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/91/819691/2/check/fc-test/3eb36ba : SUCCESS in 2h 02m 26s","accounts_in_message":[],"_revision_number":2},{"id":"6347067776424905c5dd9ca8335737186412d322","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-12-02 20:40:12.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/91/819691/4/check/dsvm-tempest-cisco-zonemanager-job/0a5cac0/ : FAILURE in 1h 04m 11s","accounts_in_message":[],"_revision_number":4},{"id":"b134b1df155b26215f6853b599a4bdfffb33f9fd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-02 20:49:14.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-91-819691-4/ns5_nfs : FAILURE in 1h 52m 26s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-91-819691-4/ns5_iscsi : FAILURE in 1h 53m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"689947f8ba3eaefc34e84f055e49cbddf75b32c7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-02 21:17:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/91/819691/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/57b0875/ : FAILURE in 44m 11s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/91/819691/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f30ef77/ : FAILURE in 38m 58s","accounts_in_message":[],"_revision_number":4},{"id":"75ad3ccdc6e14d5a9b78d72594a341413bbbe4b8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-02 21:17:48.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/91/819691/4/check/tempest-dsvm-full-focal-py3/c007606/ : FAILURE in 2h 12m 47s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/91/819691/4/check/devstack-plugin-ceph-dsvm-focal-py3/487c642/ : TIMED_OUT in 3h 10m 40s","accounts_in_message":[],"_revision_number":4},{"id":"5c6627bc3edbcf68af611bde62ba5a9355746507","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-02 21:19:39.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/91/819691/4/check/DellEMC_Unity_FC/61a2a78/DellEMC_Unity_FC/257 : FAILURE in 42m 57s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/91/819691/4/check/DellEMC_Unity_iSCSI/55061ef/DellEMC_Unity_iSCSI/259 : FAILURE in 46m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"a248d363ae69a907e0ff64622d581270392fcf7a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-02 23:40:49.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/91/819691/4/check/iscsi-driver-nimble/aad7aa8/aad7aa8.html : FAILURE in 59m 34s","accounts_in_message":[],"_revision_number":4},{"id":"bb06be3b3ce2a5b115405d12dd783b1e410328fd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-03 01:36:25.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/221fc0ce25ec4bd79880fbc23c3325e7 : FAILURE in 1h 02m 09s","accounts_in_message":[],"_revision_number":4},{"id":"d696094d25ae65ceb012151e70b9ab1dd73a133a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-03 03:42:35.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/91/819691/4/check/DellEMC_XtremIO_FC/81afe71/DellEMC_XtremIO_FC/289 : FAILURE in 1h 00m 50s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/91/819691/4/check/DellEMC_XtremIO_iSCSI/e164b82/DellEMC_XtremIO_iSCSI/304 : FAILURE in 1h 10m 39s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"e445e59e88992bf716b4bfcaad95850401eea4f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-12-03 15:07:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d2b19f2f50e039c0d33c885546ac8b780e589947","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-12-03 15:08:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"79c77e0fa6d8fe5d3c8276805d7ee48558e77029","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-03 15:22:42.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819691/5/iscsi/ : FAILURE in 13m 47s\n- cinder-smb http://cloudbase-ci.com/cinder/819691/5/smb/ : FAILURE in 13m 44s","accounts_in_message":[],"_revision_number":5},{"id":"2c31b8440e79468b0ce27f349badc12d1843283c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-03 16:05:25.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/91/819691/5/check/iscsi-driver-alletra-6k/17b810f/17b810f.html : FAILURE in 53m 39s","accounts_in_message":[],"_revision_number":5},{"id":"e4c3a421d07bfa0ee793be8513915cd651f3a000","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-03 16:09:10.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/91/819691/5/check/iscsi-driver-nimble/2b55df9/2b55df9.html : FAILURE in 58m 03s","accounts_in_message":[],"_revision_number":5},{"id":"e31e22c37d6507f9b1c25a2eb737b102c388ceff","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-12-03 16:24:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/91/819691/5/check/dsvm-tempest-cisco-zonemanager-job/947d5b6/ : SUCCESS in 1h 09m 29s","accounts_in_message":[],"_revision_number":5},{"id":"4984471bb4099454d42a7f57d2f0b53d9187cbe0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-03 16:44:11.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b47104bba8114180b9ccc5015ca8a759 : FAILURE in 1h 26m 49s","accounts_in_message":[],"_revision_number":5},{"id":"a5e4cb4c66fc2a4482ee900017c7621eb0105388","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-03 16:52:19.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/91/819691/5/check/DellEMC_Unity_FC/f3f4789/DellEMC_Unity_FC/266 : FAILURE in 1h 16m 06s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/91/819691/5/check/DellEMC_Unity_iSCSI/b8a54a3/DellEMC_Unity_iSCSI/268 : FAILURE in 1h 44m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"638703bc0888f7509fce833aee8d591904c2ee2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 17:12:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/af9ed98f709e4d46b5eb81cfcc14c327 : FAILURE in 1h 52m 59s","accounts_in_message":[],"_revision_number":5},{"id":"f7be67a927bf24d9dc865a2bd54b89f34cc422f4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-03 17:53:24.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/91/819691/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/9b1b1b2/ : FAILURE in 2h 13m 17s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/91/819691/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/cf3f1eb/ : FAILURE in 2h 43m 45s","accounts_in_message":[],"_revision_number":5},{"id":"1947b8f1d484442859f25a693170f01212f9bde6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-03 18:37:21.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/91/819691/5/check/tempest-dsvm-full-focal-py3/9a3d90e/ : FAILURE in 3h 01m 07s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/91/819691/5/check/devstack-plugin-ceph-dsvm-focal-py3/86676a0/ : TIMED_OUT in 3h 11m 33s","accounts_in_message":[],"_revision_number":5},{"id":"66e03c44193100db0269c93aa223ba3dfafc6c48","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-03 18:37:50.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/91/819691/5/check/DellEMC_XtremIO_FC/bfc8087/DellEMC_XtremIO_FC/298 : FAILURE in 1h 04m 39s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/91/819691/5/check/DellEMC_XtremIO_iSCSI/a981575/DellEMC_XtremIO_iSCSI/313 : FAILURE in 1h 50m 53s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"57379cce3d93c6ccd6ff424642d2e895984f2054","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-03 19:35:10.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-91-819691-5/ns5_nfs : FAILURE in 3h 44m 13s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-91-819691-5/ns5_iscsi : FAILURE in 3h 54m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"a9bad34a7e2fc21c0ca5c02cca8b5663986250d2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-03 19:35:50.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"e8fcf99b28b0270a357153403b0bab8864d8a2ac","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-12-05 17:11:34.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/91/819691/5/20211205195602/NEC-FC : FAILURE in 6h 11m 19s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/91/819691/5/20211205195602/NEC-ISCSI : FAILURE in 6h 15m 28s","accounts_in_message":[],"_revision_number":5},{"id":"5b1cc16471c2de0c083a2f51793a424639ef5c06","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-05 18:36:47.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/5/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"4a4205c90c269d1bba0e8849751eeb566447233c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-12-06 07:44:04.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/819691/819691/5/check-not-vote/ext-cinder-zuul/7d7cce8 : FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"967da4b793d05f413523b4c1914490c6411a756f","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-06 09:37:56.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/91/819691/5/check/dsvm-tempest-TOYOU-ACS5000-driver/6b577b4 : FAILURE in 1h 59m 14s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/91/819691/5/check/fc-test/06bec4c : FAILURE in 2h 05m 24s","accounts_in_message":[],"_revision_number":5},{"id":"79babc3b2eed82f706704d3fe80903d2a8bf0cf9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-06 10:51:10.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/5/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"63ba7ae437c73ecbe57516199c34fccb978d9b90","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-12-07 17:58:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e3080276f847b3b476b01d42a7d06ab52a54695e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-12-07 17:59:02.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"94290c831b8e7d767834f6ed470cfddd857d06f5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-07 18:11:35.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k finger://openstackcifc-virtual-machine/5cc2881754c44aafa88081b7d1701f7d : RETRY_LIMIT in 10s","accounts_in_message":[],"_revision_number":6},{"id":"d8f9f041eb2c791c99ab1c7922737c7568755b32","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-07 18:28:47.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"8bd1b677e17f67c03fc5e1eaf6c51132f1fcb843","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-07 19:03:02.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/6/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"bea286bcd3d85d7796185e03441179228e27aaa9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-07 19:06:44.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819691/6/iscsi/ : FAILURE in 1h 05m 47s\n- cinder-smb http://cloudbase-ci.com/cinder/819691/6/smb/ : FAILURE in 1h 00m 25s","accounts_in_message":[],"_revision_number":6},{"id":"a222625365e36cf102231d17f4bc692197e6fa2b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-07 19:35:17.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/6/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"8ccf01102863d5d9ba2fdc3b356cc607294c6a73","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-12-07 20:08:04.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/91/819691/6/check/dsvm-tempest-cisco-zonemanager-job/6f2a224/ : SUCCESS in 1h 09m 56s","accounts_in_message":[],"_revision_number":6},{"id":"b084c7f160df70c14f6618522cdc6aa280130143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 20:23:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/71c2481b7bbc4b27bc56c410ced5049e : FAILURE in 1h 29m 28s","accounts_in_message":[],"_revision_number":6},{"id":"c0586a7c2bd29c59f1e9f2e6adb01114089421d7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-07 20:33:22.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/91/819691/6/check/iscsi-driver-nimble/27e376f/27e376f.html : FAILURE in 1h 00m 16s","accounts_in_message":[],"_revision_number":6},{"id":"398434cbd6331b561fe2d442807b46b8df49d6b6","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-07 20:55:22.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/91/819691/6/check/DellEMC_XtremIO_FC/7c37246/DellEMC_XtremIO_FC/319 : FAILURE in 1h 04m 49s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/91/819691/6/check/DellEMC_XtremIO_iSCSI/39387e9/DellEMC_XtremIO_iSCSI/332 : FAILURE in 1h 39m 05s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":6},{"id":"d2c44c4d5ba32f793e53ba42def55f565dd64132","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-07 20:57:48.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e365c414ab3245b6b0e2bb7245fc5c72 : FAILURE in 1h 18m 12s","accounts_in_message":[],"_revision_number":6},{"id":"77ecffbd2d2489bdf0321aded07bb5b0eb8cfac4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-07 21:04:22.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/91/819691/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5ca8224/ : TIMED_OUT in 3h 02m 34s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/91/819691/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/779764e/ : TIMED_OUT in 3h 02m 58s","accounts_in_message":[],"_revision_number":6},{"id":"d142253e4f35d49aa2de56707912d3e1f5569c49","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-07 22:09:44.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/6/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"04a5fd4ccadc3ad264d6b705ea79b405dc315b77","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-07 22:11:46.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/91/819691/6/check/dsvm-tempest-TOYOU-ACS5000-driver/63bdc00 : FAILURE in 2h 17m 18s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/91/819691/6/check/fc-test/98105ac : FAILURE in 2h 04m 47s","accounts_in_message":[],"_revision_number":6},{"id":"6b48fc4b4d36a79aaf1db4203605d6c622ddb627","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-07 22:29:41.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/6/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":6},{"id":"bed46fc712bcc43d7e3632c3c5565b7e18b57279","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-07 23:09:21.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-91-819691-6/ns5_nfs : FAILURE in 4h 1m 42s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-91-819691-6/ns5_iscsi : FAILURE in 4h 1m 21s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"255a8ddd5f0e69cbf5a970c344772de9a1d486a7","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-12-08 15:06:46.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/91/819691/6/20211208173731/NEC-FC : FAILURE in 6h 23m 13s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/91/819691/6/20211208173731/NEC-ISCSI : FAILURE in 6h 29m 04s","accounts_in_message":[],"_revision_number":6},{"id":"24612fa31bf7c0df38c1b1472d52a4a72f82aa2f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-08 16:27:08.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/6/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"8c79f8223d3c75e530e52bf42e42738b43fd647e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-12-08 19:43:38.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/819691/819691/6/check-not-vote/ext-cinder-zuul/b7fedf7 : FAILURE in 35m 56s","accounts_in_message":[],"_revision_number":6},{"id":"40ebcba0393d665564f44622c7bbdab5405a6b45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-12-09 10:53:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0f02c2d42860bb69b774759e6e1987fa68fa38b3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-09 11:36:45.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"35f3ca947400ffaec274899e677b67c52f2f5218","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-09 11:54:52.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/91/819691/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a202c62/ : FAILURE in 14m 44s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/91/819691/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6da1df7/ : FAILURE in 58m 57s","accounts_in_message":[],"_revision_number":7},{"id":"c423a99ebc0e7188cd12eb0cf024a620ab16e4d0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-09 11:59:02.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819691/7/iscsi/ : FAILURE in 1h 02m 58s\n- cinder-smb http://cloudbase-ci.com/cinder/819691/7/smb/ : FAILURE in 1h 03m 41s","accounts_in_message":[],"_revision_number":7},{"id":"14054eef4f3730d8be58c509efe5683d182ef84a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-09 11:59:51.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/7/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"21eb0ab88bc2524b47c5c3a9b15211865b4c7c24","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-09 12:03:57.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/91/819691/7/check/iscsi-driver-alletra-6k/1b90e36/1b90e36.html : FAILURE in 54m 05s","accounts_in_message":[],"_revision_number":7},{"id":"02cd8e42c1335a24dfbea22dcc45e3269d4bb9bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 12:32:59.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fbf758533d9d457aa48240ad12d3ed6c : FAILURE in 1h 32m 16s","accounts_in_message":[],"_revision_number":7},{"id":"c62aa69c95514f0b9bc08a15b23c568c5f188993","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-09 13:02:03.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/91/819691/7/check/iscsi-driver-nimble/0845d23/0845d23.html : FAILURE in 59m 13s","accounts_in_message":[],"_revision_number":7},{"id":"12d7517d1bd4bf02c0e1f7e62e6f0e9cd9ba26b8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-12-09 13:17:05.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/91/819691/7/check/dsvm-tempest-cisco-zonemanager-job/4d5f928/ : SUCCESS in 1h 10m 39s","accounts_in_message":[],"_revision_number":7},{"id":"42bc1fd3fe750b8e3c2cf0d7efafd7a782a6fad7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-12-09 13:40:44.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/819691/819691/7/check-not-vote/ext-cinder-zuul/e98bf75 : FAILURE in 25s","accounts_in_message":[],"_revision_number":7},{"id":"5d820668dd0e3f57ae08a25f24a0f19da1edc667","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-12-09 13:57:06.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"25995cd7bf78da5505b6a14c42486dfdeac5d1b6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-09 14:25:09.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"2cf9b477c55d23c477e77c08b6a9a6f4a8418971","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-12-09 14:27:08.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/819691/819691/8/check-not-vote/ext-cinder-zuul/5e1d8dc : FAILURE in 25s","accounts_in_message":[],"_revision_number":8},{"id":"43f42b5652292cd2d281a43da7491b5a1fcf7039","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-09 14:51:09.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/819691/8/iscsi/ : FAILURE in 38m 01s\n- cinder-smb http://cloudbase-ci.com/cinder/819691/8/smb/ : FAILURE in 52m 09s","accounts_in_message":[],"_revision_number":8},{"id":"871cf06a349b59d6e24f214e672965690b8bbca8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-09 14:51:10.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"e5286db720715209bd46d06445a136fbf9211f32","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-09 14:57:43.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/91/819691/8/check/iscsi-driver-alletra-6k/f5206d1/f5206d1.html : FAILURE in 53m 41s","accounts_in_message":[],"_revision_number":8},{"id":"a280e56a7bcb8f47361a6c958d9f15810a3fa0e8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-09 14:57:46.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"13398ee9c9a794b44573b20b7353d937028d49b5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-09 15:02:01.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/91/819691/8/check/iscsi-driver-nimble/e0dd34e/e0dd34e.html : FAILURE in 59m 01s","accounts_in_message":[],"_revision_number":8},{"id":"f03e1e40ac89a41dd07a6a03ca936fe400e54bc7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-12-09 15:15:17.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/91/819691/8/check/dsvm-tempest-cisco-zonemanager-job/5c9ad15/ : SUCCESS in 1h 09m 09s","accounts_in_message":[],"_revision_number":8},{"id":"b2b284372896827ba4fee7ebb2a37169fa44f372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 15:39:55.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ee325b95a9b04bcfba69322f08811f60 : FAILURE in 1h 36m 08s","accounts_in_message":[],"_revision_number":8},{"id":"62527e23cfea18d25789d675e1ccec8d3038fe93","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-09 17:02:41.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/91/819691/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/544ac57/ : TIMED_OUT in 3h 02m 51s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/91/819691/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/dea1481/ : TIMED_OUT in 3h 03m 10s","accounts_in_message":[],"_revision_number":8},{"id":"ac02ae19657b04017b64e11d575773161d4cd736","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-09 17:47:59.000000000","message":"Patch Set 8:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/91/819691/8/check/dsvm-tempest-TOYOU-ACS5000-driver/d014632 : FAILURE in 2h 06m 07s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/91/819691/8/check/fc-test/de64b31 : FAILURE in 2h 21m 07s","accounts_in_message":[],"_revision_number":8},{"id":"7217cce49bde43557d83c586e7084f659fbf7842","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-09 18:08:53.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"fca0fb58628b28b90f21826ac3f424f94800a58e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-09 18:57:59.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-91-819691-7/ns5_nfs : FAILURE in 4h 12m 15s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-91-819691-7/ns5_iscsi : FAILURE in 4h 16m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"935abe98376e7a392537baa44a49ba58628eb819","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-09 19:52:53.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/7/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":7},{"id":"3b8499830f365f917d407da668fc85952c091a6b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-09 20:45:11.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-91-819691-8/ns5_nfs : FAILURE in 4h 15m 59s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-91-819691-8/ns5_iscsi : FAILURE in 4h 33m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"0bb97195e2afde750a18540a1e5f5a76e3c2aba3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-12-09 21:31:52.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/91/819691/8/20211210001502/NEC-FC : FAILURE in 6h 12m 33s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/91/819691/8/20211210001502/NEC-ISCSI : FAILURE in 6h 16m 46s","accounts_in_message":[],"_revision_number":8},{"id":"224c5022c38003021a04f4ea782f16d9f8f0f7c3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-09 21:32:51.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/8/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"8b9a72c438e040ae5855850c2b52f946f70d31f9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-09 22:43:24.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/819691/8/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"8f7603cce30dbc7533fba56fc9bd65ca35662d1c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-11 00:12:06.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f9f30cf946474f6fa89e7a78330284bb : FAILURE in 1h 27m 09s","accounts_in_message":[],"_revision_number":8},{"id":"359cc7b94b9881fe6bbcb9e37ff71b0fbfe166be","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-05-25 14:11:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"60103e35902e0237e6c640406794f44cff105196","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-01-29 22:52:59.000000000","message":"Patch Set 8: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":8},{"id":"fa441065f949b1a6ab348bf9af7d56f8276d4b16","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-01-29 22:53:34.000000000","message":"Abandoned\n\nin favor of https://review.opendev.org/c/openstack/cinder/+/907152 (since i thought doing it over master would be better, it wasn\u0027t)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"c3aa8c9e9c68db847e95f1205e4754ddb6ab4dc4","revisions":{"9d7fb2657af4c2c3e90785627f3d3f6a2f7d9deb":{"kind":"REWORK","_number":1,"created":"2021-11-29 18:19:05.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/91/819691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/819691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/819691/1"}}},"commit":{"parents":[{"commit":"373b7a404910fee52a94d1feb18ecbcd461ecc94","subject":"Quota: Config deprecation warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/373b7a404910fee52a94d1feb18ecbcd461ecc94"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-12 11:18:23.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-29 18:18:18.000000000","tz":60},"subject":"WIP: Quota: New quota system","message":"WIP: Quota: New quota system\n\nThis is a WIP patch to illustrate how it could look like if we replace\nthe old database quota system with a new system that dynamically counts\nresource consumption.\n\nThe intention is to help with the review of the spec.\n\nThe code is incomplete, as it\u0027s missing:\n\n- Manually check that all operations work as intended\n- Updating existing unit tests\n- Adding new unit tests\n- Properly formatting the code\n- Documenting the code\n- Adding mypy support\n\nAnd that is why, for the time being, the patch disables all CI jobs but\none.\n\nImplements: blueprint count-resource-to-check-quota\n\nCloses-Bug: #1952635\nCloses-Bug: #1602396\nCloses-Bug: #1785947\nCloses-Bug: #1952456\nCloses-Bug: #1576717\nCloses-Bug: #1609937\nCloses-Bug: #1435807\nChange-Id: Iedc9b8939ffcda2ca7fad0357aac97ae9a2df588\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d7fb2657af4c2c3e90785627f3d3f6a2f7d9deb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d7fb2657af4c2c3e90785627f3d3f6a2f7d9deb"}]},"branch":"refs/heads/master"},"7ca2cb58871cb4a624a23638eafdad7a366201d1":{"kind":"REWORK","_number":2,"created":"2021-11-30 17:25:30.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/91/819691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/819691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/819691/2"}}},"commit":{"parents":[{"commit":"13c0aeed804800be85f8cebc1e055151d526f6f8","subject":"Quota: Config deprecation warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/13c0aeed804800be85f8cebc1e055151d526f6f8"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-12 11:18:23.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-30 17:25:18.000000000","tz":60},"subject":"WIP: Quota: New quota system","message":"WIP: Quota: New quota system\n\nThis is a WIP patch to illustrate how it could look like if we replace\nthe old database quota system with a new system that dynamically counts\nresource consumption.\n\nThe intention is to help with the review of the spec.\n\nThe code is incomplete, as it\u0027s missing:\n\n- Manually check that all operations work as intended\n- Updating existing unit tests\n- Adding new unit tests\n- Properly formatting the code\n- Documenting the code\n- Adding mypy support\n\nAnd that is why, for the time being, the patch disables all CI jobs but\none.\n\nImplements: blueprint count-resource-to-check-quota\n\nCloses-Bug: #1952635\nCloses-Bug: #1602396\nCloses-Bug: #1785947\nCloses-Bug: #1952456\nCloses-Bug: #1576717\nCloses-Bug: #1609937\nCloses-Bug: #1435807\nChange-Id: Iedc9b8939ffcda2ca7fad0357aac97ae9a2df588\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ca2cb58871cb4a624a23638eafdad7a366201d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ca2cb58871cb4a624a23638eafdad7a366201d1"}]},"branch":"refs/heads/master"},"6435e74663bc5421bcb78b063bc1b0a5b0c572f8":{"kind":"REWORK","_number":3,"created":"2021-12-01 10:17:24.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/91/819691/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/819691/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/819691/3"}}},"commit":{"parents":[{"commit":"13c0aeed804800be85f8cebc1e055151d526f6f8","subject":"Quota: Config deprecation warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/13c0aeed804800be85f8cebc1e055151d526f6f8"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-12 11:18:23.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-12-01 10:17:14.000000000","tz":60},"subject":"WIP: Quota: New quota system","message":"WIP: Quota: New quota system\n\nThis is a WIP patch to illustrate how it could look like if we replace\nthe old database quota system with a new system that dynamically counts\nresource consumption.\n\nThe intention is to help with the review of the spec.\n\nThe code is incomplete, as it\u0027s missing:\n\n- Manually check that all operations work as intended\n- Updating existing unit tests\n- Adding new unit tests\n- Properly formatting the code\n- Documenting the code\n- Adding mypy support\n\nAnd that is why, for the time being, the patch disables all CI jobs but\none.\n\nImplements: blueprint count-resource-to-check-quota\n\nCloses-Bug: #1952635\nCloses-Bug: #1602396\nCloses-Bug: #1785947\nCloses-Bug: #1952456\nCloses-Bug: #1576717\nCloses-Bug: #1609937\nCloses-Bug: #1435807\nChange-Id: Iedc9b8939ffcda2ca7fad0357aac97ae9a2df588\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6435e74663bc5421bcb78b063bc1b0a5b0c572f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6435e74663bc5421bcb78b063bc1b0a5b0c572f8"}]},"branch":"refs/heads/master"},"5ad15b5a60db3dd56cf3a59825d92409373bef8f":{"kind":"REWORK","_number":4,"created":"2021-12-02 17:55:21.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/91/819691/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/819691/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/819691/4"}}},"commit":{"parents":[{"commit":"13c0aeed804800be85f8cebc1e055151d526f6f8","subject":"Quota: Config deprecation warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/13c0aeed804800be85f8cebc1e055151d526f6f8"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-12 11:18:23.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-12-02 17:53:15.000000000","tz":60},"subject":"WIP: Quota: New quota system","message":"WIP: Quota: New quota system\n\nThis is a WIP patch to illustrate how it could look like if we replace\nthe old database quota system with a new system that dynamically counts\nresource consumption.\n\nThe intention is to help with the review of the spec.\n\nThe code is incomplete, as it\u0027s missing:\n\n- Manually check that all operations work as intended\n- Updating existing unit tests\n- Adding new unit tests\n- Properly formatting the code\n- Documenting the code\n- Adding mypy support\n\nAnd that is why, for the time being, the patch disables all CI jobs but\none.\n\nImplements: blueprint count-resource-to-check-quota\n\nCloses-Bug: #1952635\nCloses-Bug: #1602396\nCloses-Bug: #1785947\nCloses-Bug: #1952456\nCloses-Bug: #1576717\nCloses-Bug: #1609937\nCloses-Bug: #1435807\nChange-Id: Iedc9b8939ffcda2ca7fad0357aac97ae9a2df588\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ad15b5a60db3dd56cf3a59825d92409373bef8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ad15b5a60db3dd56cf3a59825d92409373bef8f"}]},"branch":"refs/heads/master"},"45dcf9dcc5f56bdd81357cb95fb0a5a1576ce796":{"kind":"REWORK","_number":5,"created":"2021-12-03 15:07:10.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/91/819691/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/819691/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/819691/5"}}},"commit":{"parents":[{"commit":"13c0aeed804800be85f8cebc1e055151d526f6f8","subject":"Quota: Config deprecation warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/13c0aeed804800be85f8cebc1e055151d526f6f8"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-12 11:18:23.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-12-03 15:06:59.000000000","tz":60},"subject":"WIP: Quota: New quota system","message":"WIP: Quota: New quota system\n\nThis is a WIP patch to illustrate how it could look like if we replace\nthe old database quota system with a new system that dynamically counts\nresource consumption.\n\nThe intention is to help with the review of the spec.\n\nThe code is incomplete, as it\u0027s missing:\n\n- Manually check that all operations work as intended\n- Updating existing unit tests\n- Adding new unit tests\n- Properly formatting the code\n- Documenting the code\n- Adding mypy support\n\nAnd that is why, for the time being, the patch disables all CI jobs but\none.\n\nImplements: blueprint count-resource-to-check-quota\n\nCloses-Bug: #1952635\nCloses-Bug: #1602396\nCloses-Bug: #1785947\nCloses-Bug: #1952456\nCloses-Bug: #1576717\nCloses-Bug: #1609937\nCloses-Bug: #1435807\nChange-Id: Iedc9b8939ffcda2ca7fad0357aac97ae9a2df588\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/45dcf9dcc5f56bdd81357cb95fb0a5a1576ce796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/45dcf9dcc5f56bdd81357cb95fb0a5a1576ce796"}]},"branch":"refs/heads/master"},"25c0b5eda42030c6af995e9074b1b8b60b4db418":{"kind":"REWORK","_number":6,"created":"2021-12-07 17:59:02.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/91/819691/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/819691/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/819691/6"}}},"commit":{"parents":[{"commit":"13c0aeed804800be85f8cebc1e055151d526f6f8","subject":"Quota: Config deprecation warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/13c0aeed804800be85f8cebc1e055151d526f6f8"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-12 11:18:23.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-12-07 17:58:48.000000000","tz":60},"subject":"WIP: Quota: New quota system","message":"WIP: Quota: New quota system\n\nThis is a WIP patch to illustrate how it could look like if we replace\nthe old database quota system with a new system that dynamically counts\nresource consumption.\n\nThe intention is to help with the review of the spec.\n\nThe code is incomplete, as it\u0027s missing:\n\n- Manually check that all operations work as intended\n- Updating existing unit tests\n- Adding new unit tests\n- Properly formatting the code\n- Documenting the code\n- Adding mypy support\n\nAnd that is why, for the time being, the patch disables all CI jobs but\none.\n\nImplements: blueprint count-resource-to-check-quota\n\nCloses-Bug: #1952635\nCloses-Bug: #1602396\nCloses-Bug: #1785947\nCloses-Bug: #1952456\nCloses-Bug: #1576717\nCloses-Bug: #1609937\nCloses-Bug: #1435807\nChange-Id: Iedc9b8939ffcda2ca7fad0357aac97ae9a2df588\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25c0b5eda42030c6af995e9074b1b8b60b4db418"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25c0b5eda42030c6af995e9074b1b8b60b4db418"}]},"branch":"refs/heads/master"},"6caec4337b8f139ad7178a0a4b0d34ab76b37d25":{"kind":"REWORK","_number":7,"created":"2021-12-09 10:53:25.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/91/819691/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/819691/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/819691/7"}}},"commit":{"parents":[{"commit":"13c0aeed804800be85f8cebc1e055151d526f6f8","subject":"Quota: Config deprecation warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/13c0aeed804800be85f8cebc1e055151d526f6f8"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-12 11:18:23.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-12-09 10:53:17.000000000","tz":60},"subject":"WIP: Quota: New quota system","message":"WIP: Quota: New quota system\n\nThis is a WIP patch to illustrate how it could look like if we replace\nthe old database quota system with a new system that dynamically counts\nresource consumption.\n\nThe intention is to help with the review of the spec.\n\nThe code is incomplete, as it\u0027s missing:\n\n- Manually check that all operations work as intended\n- Updating existing unit tests\n- Adding new unit tests\n- Properly formatting the code\n- Documenting the code\n- Adding mypy support\n\nAnd that is why, for the time being, the patch disables all CI jobs but\none.\n\nImplements: blueprint count-resource-to-check-quota\n\nCloses-Bug: #1952635\nCloses-Bug: #1602396\nCloses-Bug: #1785947\nCloses-Bug: #1952456\nCloses-Bug: #1576717\nCloses-Bug: #1609937\nCloses-Bug: #1435807\nChange-Id: Iedc9b8939ffcda2ca7fad0357aac97ae9a2df588\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6caec4337b8f139ad7178a0a4b0d34ab76b37d25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6caec4337b8f139ad7178a0a4b0d34ab76b37d25"}]},"branch":"refs/heads/master"},"c3aa8c9e9c68db847e95f1205e4754ddb6ab4dc4":{"kind":"REWORK","_number":8,"created":"2021-12-09 13:57:06.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/91/819691/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/91/819691/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/91/819691/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/91/819691/8"}}},"commit":{"parents":[{"commit":"13c0aeed804800be85f8cebc1e055151d526f6f8","subject":"Quota: Config deprecation warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/13c0aeed804800be85f8cebc1e055151d526f6f8"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-11-12 11:18:23.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-12-09 13:56:53.000000000","tz":60},"subject":"WIP: Quota: New quota system","message":"WIP: Quota: New quota system\n\nThis is a WIP patch to illustrate how it could look like if we replace\nthe old database quota system with a new system that dynamically counts\nresource consumption.\n\nThe intention is to help with the review of the spec.\n\nThe code is incomplete, as it\u0027s missing:\n\n- Manually check that all operations work as intended\n- Updating existing unit tests\n- Adding new unit tests\n- Properly formatting the code\n- Documenting the code\n- Adding mypy support\n\nAnd that is why, for the time being, the patch disables all CI jobs but\none.\n\nImplements: blueprint count-resource-to-check-quota\n\nCloses-Bug: #1952635\nCloses-Bug: #1602396\nCloses-Bug: #1785947\nCloses-Bug: #1952456\nCloses-Bug: #1576717\nCloses-Bug: #1609937\nCloses-Bug: #1435807\nChange-Id: Iedc9b8939ffcda2ca7fad0357aac97ae9a2df588\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c3aa8c9e9c68db847e95f1205e4754ddb6ab4dc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c3aa8c9e9c68db847e95f1205e4754ddb6ab4dc4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
