)]}'
{"id":"openstack%2Fcinder~821905","triplet_id":"openstack%2Fcinder~stable%2Ftrain~I05d83d73de69c57d885e0c417e8a376f7cfb1e4f","project":"openstack/cinder","branch":"stable/train","topic":"bug/1924643","attention_set":{},"removed_from_attention_set":{"33431":{"account":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"last_update":"2023-12-13 15:21:35.000000000","reason":"Change was abandoned"},"31721":{"account":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"last_update":"2021-12-16 13:28:25.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I05d83d73de69c57d885e0c417e8a376f7cfb1e4f","subject":"NetApp ONTAP: Fix sub-clone zapi call","status":"ABANDONED","created":"2021-12-16 12:25:20.000000000","updated":"2023-12-13 15:21:35.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":821894,"cherry_pick_of_patch_set":2,"meta_rev_id":"2328002b8af0793f86861482c29379ca059f04cc","_number":821905,"virtual_id_number":821905,"owner":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-12-17 13:25:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},{"value":1,"date":"2021-12-27 06:38:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":1,"date":"2022-07-08 14:50:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"value":1,"date":"2021-12-20 13:47:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2022-02-21 19:07:40.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-16 12:25:20.000000000","updated_by":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"reviewer":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"state":"REVIEWER"},{"updated":"2021-12-16 12:53:35.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-12-16 13:20:58.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2021-12-16 16:09:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-16 16:36:01.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-16 19:35:04.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-16 22:28:58.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-12-17 08:13:34.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-17 11:38:37.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2021-12-17 12:44:01.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2021-12-17 13:37:28.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-19 09:38:50.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-20 13:47:14.000000000","updated_by":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"reviewer":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"state":"REVIEWER"},{"updated":"2021-12-27 06:38:03.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"}],"messages":[{"id":"706cf0fd92dd4064d70133381239b85b7c16780c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"date":"2021-12-16 12:25:20.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"d5bb2d3f1f55324b54be1f0f4706e626b1618fc9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-16 12:53:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/2fdded05a2974a8cbf66c20c1d845538 : RETRY_LIMIT in 7m 46s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/38c36c2727aa434e81c0d8436ad14388 : RETRY_LIMIT in 7m 46s","accounts_in_message":[],"_revision_number":1},{"id":"f9bbd605db6bdf58ca8cf73768213e64eda61b8e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-16 13:20:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/05/821905/1/check/iscsi-driver-alletra-6k/45e096e/45e096e.html : FAILURE in 53m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3232005dfde56f6df252a9f1d11d61cf0cb987e9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"date":"2021-12-16 13:28:25.000000000","message":"Abandoned\n\nwrong cherry pick","accounts_in_message":[],"_revision_number":1},{"id":"1d691fc26abd6cec90b52a04cab3072c8d479e74","tag":"autogenerated:gerrit:restore","author":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"date":"2021-12-16 13:42:21.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"955af04a92fa90e3e6c2b0eac33f9281b135d910","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"date":"2021-12-16 13:49:14.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bb7762483b50066c7d4a2cdf6850904c770b8004","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-16 15:40:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/05/821905/2/check/iscsi-driver-alletra-6k/c570d02/c570d02.html : FAILURE in 53m 57s","accounts_in_message":[],"_revision_number":2},{"id":"776bec5c9c9372e2355c518bcb7898676c2aafbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 16:09:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0f8aa2d12794eccbc02619273b21035 : SUCCESS in 10m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70155c56042d4418a9309fc5a93c222f : SUCCESS in 10m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7490d29b24e45f0ac648de7d4d9afb9 : SUCCESS in 9m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7a2a1a9c8fb4ae8bf9a4f7475f4b70d : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c105c5f4a640fe8ad9dcc4339fd585 : SUCCESS in 14m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c011204078c041789bcbba6e9edfa81a : SUCCESS in 1h 35m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/376a9f64a16046f39c45869651190135 : SUCCESS in 1h 02m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/ff8358792ac4416fa0332aff6fa847c4 : SUCCESS in 59m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c94c09632a5a4d72b4e05473cf8dfdb8 : SUCCESS in 1h 27m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af2e6a3d360f4bfe82da11471bb45eaf : SUCCESS in 53m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e99c9372fd374a4bbc2d28ad08572238 : SUCCESS in 7m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/338fba6a570b43f3a2fcc53047d2a416 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cff36b94ed28453caa699969a7435219 : SUCCESS in 7m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f64e10ad1ed14b0dae50a277b10961fd : FAILURE in 17m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/142611dd85ee4cedacc7725ff6c07052 : SUCCESS in 7m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9b92924c3fb448e395276a1d3d865e6e : SUCCESS in 1h 12m 01s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a838e41f22fd4f9ca5a5e9ae8b248703 : SUCCESS in 6m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b018474759e1416baeb67730a3d76d4c : FAILURE in 1h 32m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f8f439d2ce9f4698b30fea950ddf79c0 : SUCCESS in 1h 15m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/798f3c5a2948469d84cae998465ac733 : SUCCESS in 1h 21m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0a46b5a36e3847e3a450e20447a24f27 : SUCCESS in 1h 31m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3ad77c8aaf85447cacebed6f243ed6e4 : FAILURE in 2h 01m 09s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f0852fff68e64e48b55f34e861cb6e56 : SUCCESS in 1h 00m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4926772d468240c7a42680c15078cd0c : SUCCESS in 55m 39s","accounts_in_message":[],"_revision_number":2},{"id":"6478c1df4028ee39ce2f86a6d7e43f7f40601757","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-16 16:36:01.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-821905-1/ns5_nfs : SUCCESS in 4h 10m 3s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-821905-1/ns5_iscsi : FAILURE in 36m 29s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"9703e9bf400563bec8798f5a443afd7eedd19fe2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-16 19:19:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/f9f00e398c0f47d584dd1b7e693d406f : RETRY_LIMIT in 8m 14s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/72d53514e6db403f8dd8e0251c50117d : RETRY_LIMIT in 8m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b64aad042c709aa4ab60f97097ee1be52e0b062a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-16 19:35:04.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 tempest-dsvm-full-bionic-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-bionic-py3 devstack-plugin-ceph-dsvm-bionic-py3 : MERGER_FAILURE in 28m 10s","accounts_in_message":[],"_revision_number":2},{"id":"e2d93790f0548fd3adc76b0895887ec14ce8ed2c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-16 21:01:00.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-821905-2/ns5_nfs : SUCCESS in 4h 28m 43s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-821905-2/ns5_iscsi : SUCCESS in 4h 16m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"93994225ec49f87bd646eae7f864020db081760a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2021-12-16 21:38:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/821905/2/check/iscsi-driver-nimble/0686ac5/0686ac5.html : FAILURE in 1h 00m 46s","accounts_in_message":[],"_revision_number":2},{"id":"5d5d877a2c9ea853372ab9f8ba4dce0f338d68bf","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-16 22:28:58.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/05/821905/2/check/dsvm-tempest-TOYOU-ACS5000-driver/bf7165b : FAILURE in 8m 17s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/05/821905/2/check/fc-test/41a4408 : FAILURE in 10m 22s","accounts_in_message":[],"_revision_number":2},{"id":"8c78b68fc6ac474b5e757d082467571076cacaa2","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2021-12-17 08:13:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/05/821905/2/ : SUCCESS in 1h 46m 28s","accounts_in_message":[],"_revision_number":2},{"id":"c4c74ad696031bd35f2c0957ba159aaaaee5308f","author":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"date":"2021-12-17 11:37:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4d8f98e0762a6c7afd18cd36d653b4a10764154c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2021-12-17 11:38:37.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"51cc42276acb6d5cefa0f1dc1cd797b4b5f836c7","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2021-12-17 12:44:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/dea466d/index.html : SUCCESS in 1h 05m 32s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/04febbe/index.html : SUCCESS in 1h 04m 42s","accounts_in_message":[],"_revision_number":2},{"id":"04a0b6aa1a578848457e87d2b9740d2e1c9ae009","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 13:25:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5149e55bbd44291bdae4f4b9e8e0c32 : SUCCESS in 7m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/58ab42b7686e4a63aa700a5ff3ed6633 : SUCCESS in 9m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a633b314715f405db7a14a163c7b2774 : SUCCESS in 9m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/089223039d884f6dbb0d249f02128cf1 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6933c0bc679d47a889a2557e71f5627d : SUCCESS in 13m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cde06d27fa2740c298b56bcc1d650553 : SUCCESS in 1h 22m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/372509999db447668e4d5628f0aaa0ea : SUCCESS in 54m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/c6ec155f1914434498f09abf70b42bb2 : SUCCESS in 57m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/55bd5ac9aa904cc08a76fdc3d98c75c9 : SUCCESS in 1h 02m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9258f6150c5b4064b8015e17a932c127 : SUCCESS in 52m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f81c1dcd06f42ce88aa1a9e3663b726 : SUCCESS in 6m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dbc8baad9cd54e7082427c4ed2bea4db : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/53649a46257144a5b6618f42090c91a2 : SUCCESS in 5m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/be43ee87a8b54fe28baa126205168b30 : FAILURE in 12m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/34b32a00bf98442399a96e5e6bd8119d : SUCCESS in 5m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0d16c1da0f0742b1970bcf7f8de077d1 : SUCCESS in 1h 03m 20s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7193ae3084e942d8be23a053691a9e4c : SUCCESS in 6m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e36824c96b9343f481961a51f0a5eada : SUCCESS in 1h 28m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/52eee7f78e2d4de6883fb349edb67385 : SUCCESS in 1h 13m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c659cea92ec647228d351289f7bb9548 : SUCCESS in 1h 25m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bb02d74cf4e04625826b0c4e5830532e : SUCCESS in 1h 20m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ea2c74cf29e4e28bbe9a939ae5bc0ca : SUCCESS in 1h 43m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6b79a26447db4ef58785b08b79afdb80 : SUCCESS in 55m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44126f85fc0e41308a8447fb0c4881c4 : SUCCESS in 53m 59s","accounts_in_message":[],"_revision_number":2},{"id":"c0a03f32b3aac30c075a34b5d72697fd87a63197","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-12-17 13:37:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/821905/2/upstream-check/cinder-cDOT-iSCSI/bff15ff/ : FAILURE in 10m 32s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/821905/2/upstream-check/cinder-cDOT-NFS/ae634cc/ : FAILURE in 10m 48s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/821905/2/upstream-check/cinder-cDOT-FCP/4cad58b/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"b3ff659bcef394ec96336247ff8fbdb382d9980a","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2021-12-19 09:38:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/821905/2/20211219160412/NEC-ISCSI : SUCCESS in 2h 25m 56s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/821905/2/20211219160412/NEC-FC : SUCCESS in 2h 34m 35s","accounts_in_message":[],"_revision_number":2},{"id":"6fa3f0d497e07112f08f7264f94f441ff4023833","author":{"_account_id":32266,"name":"Venkata krishna Thumu","display_name":"VenkataKrishna","email":"venkata.krishna.reddy@ibm.com","username":"venkatakrishnathumu","status":"Active"},"date":"2021-12-20 13:47:14.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ffcbbfb5e121829a76067a4777fca054520308a6","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-12-27 06:38:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fe87715766676919d9e54c7953bdd181cfde7f56","author":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"date":"2022-01-10 20:50:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d08fe2ed6893b8956a4284748b64c502b6e1c721","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-10 21:00:22.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/05/821905/2/upstream-check/cinder-cDOT-iSCSI/888f5a4/ : FAILURE in 7m 36s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/05/821905/2/upstream-check/cinder-cDOT-NFS/e215abd/ : FAILURE in 7m 33s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/05/821905/2/upstream-check/cinder-cDOT-FCP/0b0757f/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"7cc30cf8903a493fc1717501a784df337237162e","author":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"date":"2022-02-09 17:06:58.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a73e5b2857766df72733ed40324eaed783091bd9","author":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"date":"2022-02-21 19:07:40.000000000","message":"Patch Set 2: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6068b374789daec1b4b7c1edc24aa120ea0b7160","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-07-08 14:50:13.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2328002b8af0793f86861482c29379ca059f04cc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-12-13 15:21:35.000000000","message":"Abandoned\n\nTrain is going EOL [0] and the cinder project team has decided to abandon all open patches on stable/train [1].\n[0] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/UJ7Y76LRWTZQSSDAGMPWCCVVGGIX3TOG/\n[1] https://meetings.opendev.org/meetings/cinder/2023/cinder.2023-12-13-14.00.log.html#l-66","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e523d9155d879d82b01afed6aa060d99526afac8","revisions":{"4040a74ccec87f48606ab774144dd01fdd9b7b63":{"kind":"REWORK","_number":1,"created":"2021-12-16 12:25:20.000000000","uploader":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"ref":"refs/changes/05/821905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/821905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/821905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/821905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/821905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/821905/1"}}},"commit":{"parents":[{"commit":"0fb7229979858aa1a0d33dca141a110da6e6a21e","subject":"[stable-only] Pin virtualenv and setuptools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0fb7229979858aa1a0d33dca141a110da6e6a21e"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2021-04-16 13:05:40.000000000","tz":-180},"committer":{"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","date":"2021-12-16 12:25:20.000000000","tz":0},"subject":"NetApp ONTAP: Fix sub-clone zapi call","message":"NetApp ONTAP: Fix sub-clone zapi call\n\nThe ONTAP documentation states that the `clone-create` ZAPI call\nfails with `block-ranges` and `space-reserve` parameters sent\ntogether. The sub-clone uses the `block-ranges` and is failing\nbecause of that restriction.\n\nThis patch fixes the `clone-create` operation by using exactly one\nof `block-ranges` or `space-reserve`.\n\nChange-Id: I05d83d73de69c57d885e0c417e8a376f7cfb1e4f\nCloses-Bug: #1924643\n(cherry picked from commit dd0b1076df936b9cb1111567f6b4c3b747744b72)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4040a74ccec87f48606ab774144dd01fdd9b7b63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4040a74ccec87f48606ab774144dd01fdd9b7b63"}]},"branch":"refs/heads/stable/train"},"e523d9155d879d82b01afed6aa060d99526afac8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-12-16 13:49:14.000000000","uploader":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"ref":"refs/changes/05/821905/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/821905/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/821905/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/821905/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/821905/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/821905/2"}}},"commit":{"parents":[{"commit":"0fb7229979858aa1a0d33dca141a110da6e6a21e","subject":"[stable-only] Pin virtualenv and setuptools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0fb7229979858aa1a0d33dca141a110da6e6a21e"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2021-04-16 13:05:40.000000000","tz":-180},"committer":{"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","date":"2021-12-16 13:48:57.000000000","tz":0},"subject":"NetApp ONTAP: Fix sub-clone zapi call","message":"NetApp ONTAP: Fix sub-clone zapi call\n\nThe ONTAP documentation states that the `clone-create` ZAPI call\nfails with `block-ranges` and `space-reserve` parameters sent\ntogether. The sub-clone uses the `block-ranges` and is failing\nbecause of that restriction.\n\nThis patch fixes the `clone-create` operation by using exactly one\nof `block-ranges` or `space-reserve`.\n\nChange-Id: I05d83d73de69c57d885e0c417e8a376f7cfb1e4f\nCloses-Bug: #1924643\n(cherry picked from commit dd0b1076df936b9cb1111567f6b4c3b747744b72)\n(cherry picked from commit 344f3e8b098ff52c137ece8a28b3b40a5893e30e)\n(cherry picked from commit fffe9b57b984ba402fbe9f0c578b419926491b5b)\n(cherry picked from commit 9a28bce3eefb25c02316c01995e88352f8ae9528)\n(cherry picked from commit d4dd49451b5e7979f98b4250b8a7e4f99b7f3ee4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e523d9155d879d82b01afed6aa060d99526afac8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e523d9155d879d82b01afed6aa060d99526afac8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
