)]}'
{"id":"openstack%2Fcinder~823185","triplet_id":"openstack%2Fcinder~master~Ia1d7d7605686afb68900801dff8aca7be28b97d0","project":"openstack/cinder","branch":"master","topic":"bug/1955670","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-01-13 09:10:21.000000000","reason":"Rajat Dhasmana replied on the change"}},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-01-13 09:10:21.000000000","reason":"Rajat Dhasmana replied on the change"}},"hashtags":[],"change_id":"Ia1d7d7605686afb68900801dff8aca7be28b97d0","subject":"Don\u0027t allow transfer of volume with group","status":"NEW","created":"2021-12-30 07:22:04.000000000","updated":"2022-01-15 11:05:21.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"ac19f8851e06d0488c2cf5e35d19673af16d7746","_number":823185,"virtual_id_number":823185,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-12-30 12:10:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":-1,"date":"2022-01-12 16:31:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-30 07:32:03.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-30 07:34:53.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2021-12-30 07:42:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-30 07:43:37.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2021-12-30 09:39:38.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-30 09:41:07.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-30 09:47:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-30 10:02:10.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-30 10:55:56.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2021-12-30 11:35:56.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-30 11:43:20.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-30 12:23:38.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-12-30 13:17:19.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-12-30 13:24:29.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-30 17:30:25.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-01-04 06:22:09.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-04 11:09:24.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-12 16:31:09.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-01-15 11:05:21.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"b119f4ab5fb49e81f75e9ca04cf8d687d50372e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-12-30 07:22:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c369d525f8a9da46f635abc7c0a87dd14763e602","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-30 07:32:03.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/85/823185/1/check/DellEMC_XtremIO_FC/331c629/DellEMC_XtremIO_FC/375 : FAILURE in 9m 40s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/85/823185/1/check/DellEMC_XtremIO_iSCSI/da216ba/DellEMC_XtremIO_iSCSI/389 : FAILURE in 9m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"aa399d04118264de283833162e8d1393fc0fe45d","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-30 07:34:53.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/85/823185/1/check/dsvm-tempest-TOYOU-ACS5000-driver/9f3d5d5 : FAILURE in 9m 44s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/85/823185/1/check/fc-test/30d230e : FAILURE in 11m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d3af3241ae8f58063c2262958d0760024ba405d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 07:42:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c2cca1395bc4e8e92c5f1f066b996c7 : SUCCESS in 17m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff6d2a75cc1c45ac94e14a6747a02198 : SUCCESS in 16m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ee3d3e5238bc2242f93c644c9ebb24ad754dd58","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-30 07:43:37.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/823185/1/iscsi/ : FAILURE in 20m 58s\n- cinder-smb http://cloudbase-ci.com/cinder/823185/1/smb/ : FAILURE in 20m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d2d609ea98f4ced8de0f055c2ac71c96f98de35d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-30 09:39:38.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/85/823185/1/check/tempest-dsvm-full-focal-py3/a13e213/ : SUCCESS in 1h 53m 03s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/85/823185/1/check/devstack-plugin-ceph-dsvm-focal-py3/043d612/ : SUCCESS in 2h 13m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f153e2e9175153979a6d6488f8aea670e825b80f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-30 09:41:07.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/85/823185/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"3807853a3be01b3553ded43ca0ee70333d907bf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 09:47:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/400c3e2f405d49cd903c930ba0c0ba3a : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/368b71745ac0436f913c75509f259f69 : SUCCESS in 10m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6868257df9fb40ab9fd8ba4829bea1ba : SUCCESS in 9m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bd8078b7f4d45d481cfa847265f776f : SUCCESS in 12m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/83283b50a48a4dccb4b8e26871ad49f6 : SUCCESS in 57m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/74c6a0609ff349a3964ea7ea7fd02824 : SUCCESS in 1h 02m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3068f560786640b8a503694e3b9caf00 : SUCCESS in 47m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7f04ccf414794294b73c42637582d099 : SUCCESS in 14m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/36323e72fa554ea291a93aa4e812d474 : SUCCESS in 6m 13s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5977c04fbd4541448f219ea08d70ea1c : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b4d84455d7084a8a97476bb6c0fe7cca : SUCCESS in 7m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1c3aa6136d4f4ac7bf09326f30645faf : SUCCESS in 7m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f06df897fe0c465699a10ffa9bf84820 : SUCCESS in 36m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5fb0d32379f4462d9fcc574439b127f6 : SUCCESS in 5m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0b3f322091d44b30b1eeaceec4aa2b4f : SUCCESS in 1h 22m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/983576fd4d1649b4a344bf25f0f302ac : SUCCESS in 1h 38m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f37a0f56bf7e4ebab30e4435668bca1a : SUCCESS in 1h 26m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4c07e6566f40401ba9f0fcc0817ac893 : SUCCESS in 1h 22m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/034cbaf5bce24c87b2a40f3e2be05fc5 : SUCCESS in 1h 20m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/691137866fa34f40aa316f25c046365a : SUCCESS in 40m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4f9311a0f8c7459b960aeaad224144ee : SUCCESS in 1h 18m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad6410167cd4445a9fbdbf4ed51aeac4 : SUCCESS in 2h 22m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f7a5cc14e5f471293e31fde5f017c02 : SUCCESS in 53m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e675ba8e1c848775f9223b47c18008a017985638","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-30 09:57:25.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/85/823185/1/check/DellEMC_XtremIO_FC/b334b66/DellEMC_XtremIO_FC/378 : FAILURE in 9m 55s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/85/823185/1/check/DellEMC_XtremIO_iSCSI/907e76b/DellEMC_XtremIO_iSCSI/392 : FAILURE in 9m 07s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"ca728e2b5d8c888eaaade089992ffa011e73ac46","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-12-30 10:02:10.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/85/823185/1/check/DellEMC_VNX_iSCSI/42c01bc/DellEMC_VNX_iSCSI/177 : FAILURE in 9m 55s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/85/823185/1/check/DellEMC_VNX_FC/e786574/DellEMC_VNX_FC/184 : FAILURE in 14m 38s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"15f198fc43a6e0e18d19e791a01a9eff430fd8f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-12-30 10:09:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c063687213660c1a2508040556d588f6ca15d1c4","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-30 10:19:24.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/85/823185/2/check/DellEMC_XtremIO_FC/7f5a4be/DellEMC_XtremIO_FC/379 : FAILURE in 9m 32s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/85/823185/2/check/DellEMC_XtremIO_iSCSI/2754330/DellEMC_XtremIO_iSCSI/393 : FAILURE in 9m 20s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"3770222033ca8faa3005b43f2ef37c1df1741d79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 10:30:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/95495e7a13b348058cccb3afccc3ac34 : SUCCESS in 14m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9036e00c36054fcf8e628598deda21df : SUCCESS in 16m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"edd2e23266ebd5e31fcbe2af814025f37b449b3b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2021-12-30 10:32:46.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/823185/2/iscsi/ : FAILURE in 21m 24s\n- cinder-smb http://cloudbase-ci.com/cinder/823185/2/smb/ : FAILURE in 22m 49s","accounts_in_message":[],"_revision_number":2},{"id":"e7f31016c6536d0362b3b8c6a3912f456ce1c7f2","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-12-30 10:55:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n- Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/823185-1/Seagate-iSCSI : FAILURE in 12:36.06\n- Seagate-FC http://os-logs.tristero.net/seagate-ci/823185-1/Seagate-FC : FAILURE in 12:28.97","accounts_in_message":[],"_revision_number":1},{"id":"cd2deaf65fadc593ef481779efb95ffdc8fa0686","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-12-30 11:35:56.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/364a6aa9fcda41e298cb11d8f70569e0 : FAILURE in 1h 23m 25s","accounts_in_message":[],"_revision_number":2},{"id":"1500e4e96b019acf7a991757a157f2248507e5ef","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-30 11:43:20.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-85-823185-1/ns5_nfs : SUCCESS in 4h 20m 40s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-85-823185-1/ns5_iscsi : SUCCESS in 4h 14m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"71c8a49f67fd1d9e3d03dda92b5b766dcd6081d2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-12-30 11:50:57.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/85/823185/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"e23e2e641a2078615e6c01f9cd08b5d7675ef516","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-30 12:10:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1220561446e243f19534a6fd7016cb25 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13d03cf4ad5244eab072a519f5dfb130 : SUCCESS in 11m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7cbfc9c17d02417ebfd8279ca336fe0d : SUCCESS in 11m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bc937ef1b174ffea0a66c478af31351 : SUCCESS in 13m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/e810221db4684395bda1fc0f0caeb440 : SUCCESS in 55m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/956ecd496e8e4caea9535642a00b496d : SUCCESS in 1h 08m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e33ec3457084f659485ae0d853ab8d4 : SUCCESS in 45m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08a95e425ed942fd85d1881a8dfade1f : SUCCESS in 6m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/190f94ff309248a4bfa912b47a15bb8d : SUCCESS in 13m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4c33186d84f14ca888bc59ba27195a73 : SUCCESS in 7m 19s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/267aa6b6150045d2b683892c3071c304 : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/85441b90214c4a0a8fbae339a695aae1 : SUCCESS in 8m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a9816ad9e02547a48e2a3d5b4c4463bd : SUCCESS in 8m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/37392cbd89a843509a6e9a93cbb0e512 : SUCCESS in 37m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8fa9a602b96e42ecaba00161ff5fbe94 : SUCCESS in 7m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2324eb09d29149b8a1bdfaf1b3fdea9c : SUCCESS in 1h 19m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9b731b38f4aa45bf8d1594843b91d339 : SUCCESS in 1h 19m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/43aee18560ab43d5991db2c3c3dac819 : SUCCESS in 1h 42m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7e1aa728e5c4464096a8fc4cd9dd5a2a : SUCCESS in 1h 07m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0c7b5a4df22e434db9d68dc25c7e1c83 : SUCCESS in 1h 13m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7e3aa2a9ed3d41b190a970f4ad14ad6c : SUCCESS in 40m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/03c06567af3b4fe88f787affc15964f3 : SUCCESS in 1h 19m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/24d30e10432e43149bf5d494b433c79d : SUCCESS in 1h 57m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b465993c01c543c4b276ddd64dfc6e35 : SUCCESS in 38m 31s","accounts_in_message":[],"_revision_number":2},{"id":"9305a90760ff7481a5579053a5e1241fc1aec41a","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-12-30 12:23:38.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/85/823185/2/check/DellEMC_SC_FC/77d7b38/DellEMC_SC_FC/154 : FAILURE in 7m 57s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/85/823185/2/check/DellEMC_SC_iSCSI/2f0144a/DellEMC_SC_iSCSI/151 : FAILURE in 12m 50s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"17af3713e25874f0be9ca3eef2d798fd29ff275c","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-12-30 12:24:26.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/85/823185/2/check/DellEMC_XtremIO_FC/9a9f57f/DellEMC_XtremIO_FC/382 : FAILURE in 9m 38s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/85/823185/2/check/DellEMC_XtremIO_iSCSI/29122ce/DellEMC_XtremIO_iSCSI/396 : FAILURE in 9m 14s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"f668db37ee1c7c41863bc98a5e16c30d991c9d28","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-12-30 12:25:32.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/85/823185/2/check/DellEMC_VNX_iSCSI/b9106a9/DellEMC_VNX_iSCSI/178 : FAILURE in 9m 43s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/85/823185/2/check/DellEMC_VNX_FC/6dfc54f/DellEMC_VNX_FC/185 : FAILURE in 14m 41s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"007e21891671944408e8661218f61071d8a224fc","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2021-12-30 13:05:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n- Seagate-iSCSI http://os-logs.tristero.net/seagate-ci/823185-2/Seagate-iSCSI : FAILURE in 12:23.97\n- Seagate-FC http://os-logs.tristero.net/seagate-ci/823185-2/Seagate-FC : FAILURE in 12:34.94","accounts_in_message":[],"_revision_number":2},{"id":"2067f57c837715c1f40a84006ab5109721c41399","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-12-30 13:17:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/85/823185/2/upstream-solidfire-check/cinder-solidfire-iSCSI/4e208bb/ : SUCCESS in 1h 05m 27s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"0d2de3f5644a462a53421acf40013823ada0d480","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-12-30 13:24:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/85/823185/2/upstream-check/cinder-cDOT-iSCSI/2a4bc4a/ : SUCCESS in 34m 38s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/85/823185/2/upstream-check/cinder-cDOT-NFS/f5dcf96/ : SUCCESS in 35m 07s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/85/823185/2/upstream-check/cinder-cDOT-FCP/ca86565/ : FAILURE in 1h 12m 54s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"4a0273e8dc02899bde20a4dee8a4614b5a9bebdd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-12-30 13:26:51.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/85/823185/2/check/tempest-dsvm-full-focal-py3/058fc8e/ : SUCCESS in 1h 57m 39s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/85/823185/2/check/devstack-plugin-ceph-dsvm-focal-py3/73f8e62/ : TIMED_OUT in 3h 12m 21s","accounts_in_message":[],"_revision_number":2},{"id":"fba54401f0c145916a7fab4338d33da32ffe168e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-30 14:22:16.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-85-823185-2/ns5_nfs : SUCCESS in 4h 10m 45s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-85-823185-2/ns5_iscsi : SUCCESS in 4h 12m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"08cb2df435fc5ef788e5a9576f7131802be3e494","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-30 17:30:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/a5f383bb8cf449aa9703149995ded8fc : RETRY_LIMIT in 1h 52m 21s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/c2f4fef7070243ee87453f922941af49 : RETRY_LIMIT in 1h 01m 26s","accounts_in_message":[],"_revision_number":2},{"id":"0e12afe5462245c89351a982513e9fd5301a70c1","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2021-12-31 06:06:35.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/85/823185/2/check/dsvm-tempest-TOYOU-ACS5000-driver/3366c59 : SUCCESS in 3h 03m 17s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/85/823185/2/check/fc-test/e68f0ed : FAILURE in 2h 36m 06s","accounts_in_message":[],"_revision_number":2},{"id":"59b6eda91b95cf220684be7f1a66594f8cdd1b60","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-04 06:22:09.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/823185/823185/2/check-not-vote/ext-cinder-zuul/b595a9c : FAILURE in 19m 26s","accounts_in_message":[],"_revision_number":2},{"id":"bcd50df3586cceb561b9a120552db6d2745a6902","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-01-04 11:09:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/85/823185/2/20220104171835/NEC-ISCSI : SUCCESS in 2h 46m 18s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/85/823185/2/20220104171835/NEC-FC : SUCCESS in 2h 50m 45s","accounts_in_message":[],"_revision_number":2},{"id":"f4e721841e71f2ebba7e735ec1e256af032d0fca","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-05 05:22:04.000000000","message":"Patch Set 2:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":2},{"id":"c6a0aa9a67e3325904df0503ee2a7a339875217b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-05 05:22:09.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"fb48d95971b9f9997ac2458a5aa1119e6c132b08","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-01-12 16:31:09.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b0b1dcbc48ff5fb0f9ea2489d3550db748073cf8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-01-13 09:10:21.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ac19f8851e06d0488c2cf5e35d19673af16d7746","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-01-15 11:05:21.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/85/823185/2/check/DellEMC_PowerMAX_FC/86d280b/DellEMC_PowerMAX_FC/141 : FAILURE in 1h 23m 10s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/85/823185/2/check/DellEMC_PowerMAX_iSCSI/5284633/DellEMC_PowerMAX_iSCSI/148 : FAILURE in 8m 28s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4468693ad02cd9e353e3da27ae6ea12f13909fd5","revisions":{"80cdc5b3c1f9d9e04cb22567793852349b09ea9d":{"kind":"REWORK","_number":1,"created":"2021-12-30 07:22:04.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/85/823185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/823185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/823185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/823185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/823185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/823185/1"}}},"commit":{"parents":[{"commit":"de6d54108b4152beee952ff5789861811dbe0016","subject":"Merge \"Use \u0027visibility\u0027 instead of \u0027is_public\u0027 when only use Glance API v2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de6d54108b4152beee952ff5789861811dbe0016"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-12-30 07:04:46.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-12-30 07:12:00.000000000","tz":0},"subject":"Don\u0027t allow transfer of volume with group","message":"Don\u0027t allow transfer of volume with group\n\nVolumes that are part of a group should not be allowed to be\ntransferred to a different project since the group (and other\nvolumes in that group) are bound to the source project and\ncan\u0027t (and shouldn\u0027t) be transferred with the volume.\nThe current code allows the above scenario since it only checks\nfor consistencygroup_id and not for group_id and the volume\nbecomes unusable after the transfer as any operation performed\non the volume tries to find the group which doesn\u0027t exist in\nthe destination project and errors out.\nThis patch adds a check for group_id while accepting a transfer.\n\nCloses-Bug: #1955670\nChange-Id: Ia1d7d7605686afb68900801dff8aca7be28b97d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/80cdc5b3c1f9d9e04cb22567793852349b09ea9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/80cdc5b3c1f9d9e04cb22567793852349b09ea9d"}]},"branch":"refs/heads/master"},"4468693ad02cd9e353e3da27ae6ea12f13909fd5":{"kind":"REWORK","_number":2,"created":"2021-12-30 10:09:29.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/85/823185/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/823185/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/823185/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/823185/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/823185/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/823185/2"}}},"commit":{"parents":[{"commit":"de6d54108b4152beee952ff5789861811dbe0016","subject":"Merge \"Use \u0027visibility\u0027 instead of \u0027is_public\u0027 when only use Glance API v2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de6d54108b4152beee952ff5789861811dbe0016"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-12-30 07:04:46.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-12-30 10:08:18.000000000","tz":-300},"subject":"Don\u0027t allow transfer of volume with group","message":"Don\u0027t allow transfer of volume with group\n\nVolumes that are part of a group should not be allowed to be\ntransferred to a different project since the group (and other\nvolumes in that group) are bound to the source project and\ncan\u0027t (and shouldn\u0027t) be transferred with the volume.\nThe current code allows the above scenario since it only checks\nfor consistencygroup_id and not for group_id and the volume\nbecomes unusable after the transfer as any operation performed\non the volume tries to find the group which doesn\u0027t exist in\nthe destination project and errors out.\nThis patch adds a check for group_id while accepting a transfer.\n\nCloses-Bug: #1955670\nChange-Id: Ia1d7d7605686afb68900801dff8aca7be28b97d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4468693ad02cd9e353e3da27ae6ea12f13909fd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4468693ad02cd9e353e3da27ae6ea12f13909fd5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
