)]}'
{"id":"openstack%2Fcinder~823227","triplet_id":"openstack%2Fcinder~master~Ief8858c9f236f28388d77902df41eba698e503fe","project":"openstack/cinder","branch":"master","topic":"rootwrap","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-19 02:54:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ief8858c9f236f28388d77902df41eba698e503fe","subject":"Sync rootwrap.conf from oslo.rootwrap","status":"MERGED","created":"2022-01-02 12:21:21.000000000","updated":"2022-03-19 02:56:37.000000000","submitted":"2022-03-19 02:54:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823227-rootwrap","meta_rev_id":"9d09a93808b308a9c340c6e99f64b15c38394a5f","_number":823227,"virtual_id_number":823227,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-19 02:54:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-03-18 17:07:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-01-21 14:59:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-19 01:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":1,"date":"2022-01-21 15:39:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-19 01:41:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-02 12:46:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-02 13:16:52.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-01-02 13:49:39.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-02 14:18:22.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-02 15:07:59.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2022-01-02 15:23:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-02 15:35:21.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-02 16:32:23.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-02 16:51:56.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-02 16:54:40.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-01-02 17:03:14.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-02 17:04:54.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-01-02 17:30:28.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-02 17:45:06.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-01-02 19:38:24.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-01-05 02:33:35.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-05 05:25:45.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-15 16:43:50.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-21 14:59:19.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2022-01-21 15:39:16.000000000","updated_by":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"reviewer":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"state":"REVIEWER"},{"updated":"2022-01-24 10:26:14.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"CC"},{"updated":"2022-03-18 17:07:11.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-03-19 01:41:02.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"}],"messages":[{"id":"94c4eaa640b39b1caba3bd0d7a5ba77830b91788","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-02 12:21:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9bf5e01af1f340125ce3e3d4c74043a48e74468","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-02 12:46:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd750986521c4c9492279817851557f0 : SUCCESS in 23m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cc188c76d4bf418a9b5d4c918f967878 : SUCCESS in 17m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"178524e267d09d43dcb0cacde01071bc7ebdc6e4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-02 13:16:52.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/27/823227/1/check/DellEMC_Unity_FC/48c8f97/DellEMC_Unity_FC/304 : FAILURE in 42m 34s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/27/823227/1/check/DellEMC_Unity_iSCSI/6ebc33f/DellEMC_Unity_iSCSI/302 : FAILURE in 55m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"87d6a908cbeaeb9d9305cbccdc5d5a28b775bb66","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-01-02 13:49:39.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f24751b8bbb34fe3b32632e17ea36159 : FAILURE in 1h 26m 51s","accounts_in_message":[],"_revision_number":1},{"id":"52a83a45df995dcbe1bef5ce09610852762cfb2c","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-01-02 14:18:22.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/27/823227/1/check/DellEMC_XtremIO_FC/0881c0d/DellEMC_XtremIO_FC/388 : FAILURE in 1h 56m 41s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/27/823227/1/check/DellEMC_XtremIO_iSCSI/439567c/DellEMC_XtremIO_iSCSI/402 : SUCCESS in 1h 17m 28s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"c4a8623beac65b85b237e02f4cd0d1cd2d609cbc","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2022-01-02 15:07:59.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/27/823227/1/check/dsvm-tempest-TOYOU-ACS5000-driver/c56dd30 : SUCCESS in 2h 45m 29s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/27/823227/1/check/fc-test/05b9982 : FAILURE in 2h 23m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f2df1ad9010084f051645ef08295287614846685","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-02 15:23:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bddbfb852044008b5cea751da479623 : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4adbe74826be4546867df8a2c613718d : SUCCESS in 11m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fb4b18af800448586358602b9677bdc : SUCCESS in 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff588dd2b8b340cd8848bd3eaac36098 : SUCCESS in 12m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/93d477b4d3584425aa94b606f741fcc1 : SUCCESS in 54m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/880798da45294ad2b19d58d19fc32d0f : SUCCESS in 1h 14m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b8d9364fffd495bae477e57832176b3 : SUCCESS in 44m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/91b9985aa04c4696a0b2174f3985a4f8 : SUCCESS in 14m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/185c33155c764d8594ebf418d93d0178 : SUCCESS in 6m 55s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0d51d976d50c4199b19533a8d0151e18 : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0a0930afcb404c8f92a39edf596e34e2 : SUCCESS in 7m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/085881ac9cf34791a39d062dd8d1629d : SUCCESS in 7m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a121c2d66fb24dd8a5a057a056a1efb2 : SUCCESS in 38m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/47d6c2b8aec342aea0949b70453b2639 : SUCCESS in 6m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1d69d8c2e2ef42868d03fc52fe5edf4f : SUCCESS in 1h 19m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ed9e38fb0b0b480e934777240f71eee3 : SUCCESS in 1h 37m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/63826c7754274d8d822bb3c139478541 : SUCCESS in 2h 55m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/565e94bb8d964c45b144bd046dde6ec3 : SUCCESS in 1h 13m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fb7985219c604f74bff2834515865c8b : SUCCESS in 1h 21m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/690f8f50af7340a8829122488cae6550 : FAILURE in 47m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d7540a2cd7f74f74b5018a75d23f3307 : SUCCESS in 2h 23m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/309f45ffe0dd46d4b77fa041daa05e43 : SUCCESS in 2h 23m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3df64b222364bf8b9229c40d2e76719 : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"274b0bf13a792f68a5296e5a39fda7ac6a23d065","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-02 15:35:21.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/27/823227/1/check/tempest-dsvm-full-focal-py3/90fcdee/ : TIMED_OUT in 3h 09m 45s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/27/823227/1/check/devstack-plugin-ceph-dsvm-focal-py3/56ce4b3/ : TIMED_OUT in 3h 08m 56s","accounts_in_message":[],"_revision_number":1},{"id":"6eff2288d6626d9256a3a923d459093415d7f209","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-02 16:17:04.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/27/823227/1/check/DellEMC_Unity_FC/58ecf2d/DellEMC_Unity_FC/305 : FAILURE in 45m 19s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/27/823227/1/check/DellEMC_Unity_iSCSI/3590920/DellEMC_Unity_iSCSI/303 : FAILURE in 53m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"954dd611267e172bab2dad8c124ae9a08cecaaaf","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-02 16:32:23.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-823227-1/ns5_nfs : SUCCESS in 4h 8m 56s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-27-823227-1/ns5_iscsi : SUCCESS in 4h 10m 16s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"6513749804ab5805a9ae296613b10fce5da4af2b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-01-02 16:47:38.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/27/823227/1/check/DellEMC_XtremIO_FC/0014b4c/DellEMC_XtremIO_FC/389 : SUCCESS in 55m 02s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/27/823227/1/check/DellEMC_XtremIO_iSCSI/1e042fd/DellEMC_XtremIO_iSCSI/403 : SUCCESS in 1h 24m 01s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"8336c2a3b9b8459e14a95e19b47700605327a869","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-02 16:51:56.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/27/823227/1/check/DellEMC_PowerFlex/f65c023/DellEMC_PowerFlex/220 : SUCCESS in 1h 28m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"1633d3d81a3e0b8a0a231ff0a14df58189ea4e1e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-01-02 16:54:40.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/27/823227/1/check/DellEMC_PowerStore_iSCSI/6bb2265/DellEMC_PowerStore_iSCSI/192 : FAILURE in 1h 31m 02s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/27/823227/1/check/DellEMC_PowerStore_FC/c4d9966/DellEMC_PowerStore_FC/196 : SUCCESS in 56m 15s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"de6e4836514876677fbdd7376a4f20a02dd0fd6a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-02 17:03:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/27/823227/1/upstream-check/cinder-cDOT-iSCSI/f9f5a6f/ : SUCCESS in 31m 42s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/27/823227/1/upstream-check/cinder-cDOT-NFS/af8975f/ : SUCCESS in 1h 38m 53s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/27/823227/1/upstream-check/cinder-cDOT-FCP/cb3648f/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"867843f26ce695e91264503f1f705a532643b6d5","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-01-02 17:04:54.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/27/823227/1/check/DellEMC_SC_FC/591dd55/DellEMC_SC_FC/159 : SUCCESS in 41m 42s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/27/823227/1/check/DellEMC_SC_iSCSI/a028f28/DellEMC_SC_iSCSI/156 : SUCCESS in 1h 41m 19s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"a2ca46c97043f4bde12a4c307c737b508c4f9ebd","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-01-02 17:30:28.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/27/823227/1/check/DellEMC_VNX_iSCSI/6c744f0/DellEMC_VNX_iSCSI/183 : SUCCESS in 2h 06m 51s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/27/823227/1/check/DellEMC_VNX_FC/ef88398/DellEMC_VNX_FC/190 : FAILURE in 15m 28s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"7a7317db33fe86f6541e8ae55fcefce8b4adc960","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-01-02 17:45:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/27/823227/1/upstream-solidfire-check/cinder-solidfire-iSCSI/527146b/ : SUCCESS in 2h 20m 27s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"e97aed9f1e1222c2fd8300919b03c9410e031bc2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-01-02 19:38:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/7ace1c0aa2844fdf94a3dac96423f752 : RETRY_LIMIT in 2h 55m 27s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/4886fa67f2f3488daa39a676ee6ba3ce : RETRY_LIMIT in 3h 06m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6364f6eb9ed12b6806753b5f1e3faebd69e9c27c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-05 02:33:35.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"c732a8f9fd7b6d4d3316a5b5fb1472ec06408f69","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-05 04:43:30.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/823227/823227/1/check-not-vote/ext-cinder-zuul/9262646 : FAILURE in 18m 41s","accounts_in_message":[],"_revision_number":1},{"id":"162037e9cc331954a72720e4e15a32dc30e15bde","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-01-05 05:25:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/27/823227/1/20220105112616/NEC-ISCSI : SUCCESS in 2h 40m 36s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/27/823227/1/20220105112616/NEC-FC : SUCCESS in 2h 59m 26s","accounts_in_message":[],"_revision_number":1},{"id":"1e5f7475d2609d72d82ca5e94092677c7e469d14","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-01-15 16:43:50.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/27/823227/1/check/DellEMC_PowerMAX_FC/00a421a/DellEMC_PowerMAX_FC/145 : FAILURE in 1h 23m 17s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/27/823227/1/check/DellEMC_PowerMAX_iSCSI/4f191a8/DellEMC_PowerMAX_iSCSI/152 : FAILURE in 2h 36m 21s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"da17773eb5a445586d9f8c1a3f87e428c22f7e2b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2022-01-21 14:59:19.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e04ea915daa72e63da174a8f8aba914b599d2f6a","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2022-01-21 15:39:16.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ad8bd9873c0ba74b527630ff4ce72a071c601918","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-01-24 10:26:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3613902474b69e036c4dbb7601082566f25173f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-18 17:07:11.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"92ac5df815e7c988b14976fd4f801a785705ff45","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2022-03-19 01:41:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"24a7375229293feb596994d651d8206989f278a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 01:41:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"55e29c78a341982727b5a714a09f42a74a3850ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 02:54:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d913b11b91e4bfeb6c226e51925f8b5 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0870363052cd4019979ce6ec970220e0 : SUCCESS in 12m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20d63175b2214c53ae1f1c600a5b8060 : SUCCESS in 13m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1cc94c4f3d74dfeb069d56635989876 : SUCCESS in 12m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/0e7a45a272614462b6e093757a13ad43 : SUCCESS in 55m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4d1bd2d77b9f41ff94a92e5dc6747718 : SUCCESS in 1h 12m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce4335035dc04e308a7d15243c08f307 : SUCCESS in 49m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2a4ee10179c749488b5d855da3f588ba : SUCCESS in 1h 03m 24s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f2d163d215394027a8bdb62e315e027f : SUCCESS in 1h 11m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/391bdcc28f0f444d917758304aaccd71 : SUCCESS in 51m 00s","accounts_in_message":[],"_revision_number":1},{"id":"8bfce2805869555990c59efdcad341b614c618a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 02:54:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9d09a93808b308a9c340c6e99f64b15c38394a5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 02:56:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20d2426c69714e49be267f43cfe8395d : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"35acd2b0a564655786fa18938227e9e556e784a7","revisions":{"35acd2b0a564655786fa18938227e9e556e784a7":{"kind":"REWORK","_number":1,"created":"2022-01-02 12:21:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/823227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/823227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/823227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/823227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/823227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/823227/1"}}},"commit":{"parents":[{"commit":"e5ef39604cb025f341d1ff4ea92d46a2caeb8108","subject":"Improve no_snapshot_gb_quota description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e5ef39604cb025f341d1ff4ea92d46a2caeb8108"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-02 12:20:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-02 12:20:22.000000000","tz":540},"subject":"Sync rootwrap.conf from oslo.rootwrap","message":"Sync rootwrap.conf from oslo.rootwrap\n\nThe current rootwrap.conf file is outdated and doesn\u0027t include some\nparameters. This change updates the content to make it consistent with\nthe latest example file in oslo.rootwrap.\n\nChange-Id: Ief8858c9f236f28388d77902df41eba698e503fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/35acd2b0a564655786fa18938227e9e556e784a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/35acd2b0a564655786fa18938227e9e556e784a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
