)]}'
{"id":"openstack%2Fcinder~824158","triplet_id":"openstack%2Fcinder~stable%2Fxena~Ibb77b8b1be6c35c5db3b07fdc4056afd51d48782","project":"openstack/cinder","branch":"stable/xena","topic":"bug/1935688-stable/xena","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-01-12 15:57:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibb77b8b1be6c35c5db3b07fdc4056afd51d48782","subject":"Reject bad img formats for uploaded encrypted vols","status":"MERGED","created":"2022-01-11 18:09:57.000000000","updated":"2022-01-12 16:09:50.000000000","submitted":"2022-01-12 15:57:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824158-bug/1935688-stable/xena","cherry_pick_of_change":800272,"cherry_pick_of_patch_set":2,"meta_rev_id":"a3cfa526b4c249f9b8cd85c398a066c13d08e2b0","_number":824158,"virtual_id_number":824158,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-12 15:57:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-12 14:22:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-12 14:22:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-11 18:10:10.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-01-11 18:11:40.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-01-11 18:36:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-11 19:19:20.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-01-11 19:53:17.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-11 19:53:56.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-11 20:10:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-11 20:44:28.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-11 20:51:36.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2022-01-11 21:41:46.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-01-12 14:22:50.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"d378d33bc7c27c1a2163795d160cd2f0c5adf551","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-01-11 18:09:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4b2cefef52587884809d68ea251dcfa6b95c431f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-11 18:10:10.000000000","message":"Patch Set 1:\n\nBuild started","accounts_in_message":[],"_revision_number":1},{"id":"2d1561c8902c4f3c2cec71d3bdd225c785ca25de","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-01-11 18:11:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"2bfe39d9cc6305255f29b4fc7a23562b93e457df","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-11 18:13:29.000000000","message":"Patch Set 1:\n\nBuild SUCCESS. log path: https://raw.githubusercontent.com/LightBitsLabs/logs/main/11_01_2022_18:13:28/run.log","accounts_in_message":[],"_revision_number":1},{"id":"a3a25666ca6874218dcb676debc56ba2b906d5d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 18:36:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0582f632016b44d1a450fe5b2592113b : SUCCESS in 20m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7cfb01e80bc64201bc2a629baab29689 : SUCCESS in 20m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85cf823071b07c38d51f58cf0e32728378320a6f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-11 19:19:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/824158/1/iscsi/ : SUCCESS in 1h 08m 50s\n- cinder-smb http://cloudbase-ci.com/cinder/824158/1/smb/ : SUCCESS in 1h 06m 20s","accounts_in_message":[],"_revision_number":1},{"id":"03de3f909f23a9804e9d708e93ac8c2b47c55577","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-11 19:53:17.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-824158-1/ns5_nfs : FAILURE in 1h 10m 8s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-58-824158-1/ns5_iscsi : FAILURE in 1h 9m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"fd1d3315aa316394f985333058cd2a2e6e67690f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-01-11 19:53:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/58/824158/1/ : SUCCESS in 1h 35m 48s","accounts_in_message":[],"_revision_number":1},{"id":"eb06bd3bccf6117abd6d5c09f1e24de30fcd87ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 20:10:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa3eecc2a894bf687baeb78f3e12a64 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08ae272c68384b8ab1356f3c37ec1477 : SUCCESS in 11m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea2920fb295a4e5db64002ad9ad6170a : SUCCESS in 14m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23b3df17dc4c4b9eb6e7376cf62b5dfa : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4b7cb37ab6c40ef95d0e45c20d1b1c1 : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/19169d56b1e74991afc7cc4ff51b1c22 : SUCCESS in 1h 04m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3e83281470144dffb7325444ab2355be : SUCCESS in 1h 13m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f239eceea164fcba18a3f9c89afd907 : SUCCESS in 46m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eca4ece201744274a4fc2d1f532e098e : SUCCESS in 7m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/24d07b8f6a234548b259494c15ff7cd2 : SUCCESS in 13m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/11cc52d5020547428c7115afb42198ea : FAILURE in 5m 48s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4b421f6525424e07babe0e3eb181e26c : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c554c1e8ba644d6bf8ec0478f15f882 : SUCCESS in 6m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1409424a85074a058b971d4aad5cb4d7 : SUCCESS in 6m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bd50f410b746423bb9c3044db94777e8 : SUCCESS in 41m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dc56d4c781c84588a61725ca3f67e1e9 : SUCCESS in 5m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e388ecb83d574d9b9bc603fe9838e9f4 : SUCCESS in 1h 26m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8dd6a4f4b6574600b9b5ed275bcdbbec : SUCCESS in 1h 20m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/28878f92f6d74493ac6a00c16269b976 : SUCCESS in 1h 21m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/369ac1644ae041deb86fe48a51a68523 : SUCCESS in 1h 13m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/77245399eafe4dc4ba4e3020114d484c : SUCCESS in 1h 19m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/64555eee34224e20818022beecff440c : SUCCESS in 32m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a38e0b39f8ca41ab874a4b5083baa639 : SUCCESS in 1h 12m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dcbc64b1245b4227b8603f0b00fbfc70 : SUCCESS in 1h 56m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ceb0872ff2b6496eb04f20a5be751155 : SUCCESS in 44m 15s\n\nWarning:\n  Comments left for invalid file cinder/volume/manager.py","accounts_in_message":[],"_revision_number":1},{"id":"ba0cf3b9175106d14cf7d03d46b69d8989e7d04c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-11 20:44:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/58/824158/1/upstream-check/cinder-cDOT-iSCSI/93e9e19/ : SUCCESS in 32m 45s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/58/824158/1/upstream-check/cinder-cDOT-NFS/7e0f706/ : SUCCESS in 31m 37s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/58/824158/1/upstream-check/cinder-cDOT-FCP/63f261d/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c8edc2dee1983a06b8b326b6dd933ac25dcf1d7e","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2022-01-11 20:51:36.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/58/824158/1/check/dsvm-tempest-TOYOU-ACS5000-driver/c4ff522 : SUCCESS in 2h 01m 26s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/58/824158/1/check/fc-test/694bb91 : FAILURE in 2h 08m 08s","accounts_in_message":[],"_revision_number":1},{"id":"f6b60cdb4ee5c0ab7b167e0a9c2e3cab7f66b3bb","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-01-11 21:41:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/58/824158/1/upstream-solidfire-check/cinder-solidfire-iSCSI/ee5c8a2/ : FAILURE in 1h 29m 33s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"4f3539709cd58d1c97ba40e706530265d10dce6d","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-01-12 14:22:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52ef8c8bd41d72a40e96caef3cb402a5ca14583a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 14:23:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7f5c2928c46b3303160a91b4b6eff702ef61016","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 15:57:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3ea3197494243589816463f62a6bc2d : SUCCESS in 9m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fe8ad3f2efc4b4d8bd7b3ac35f9eac8 : SUCCESS in 11m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6a511d0d22f4db0986fd704cfc0d56b : SUCCESS in 13m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61c820f16ce14fc6a8dd72aa657a8897 : SUCCESS in 12m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/927f3eabb8294f63a9c565f6960e4a88 : SUCCESS in 1h 04m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fe04a9560f354e0abca24d8beb5bd153 : SUCCESS in 1h 08m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9e709e1fac24b2686712f257b5143b9 : SUCCESS in 50m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5563dbbe9f8e45938b2090833ba5e8da : SUCCESS in 8m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fedd3f1650224e26930cc9b1579c891b : SUCCESS in 1h 16m 19s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5d09187c319b4d249abc79321c4aae78 : SUCCESS in 1h 23m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6f6416971ca4d5e99bcd6f4e6bf2b16 : SUCCESS in 52m 40s","accounts_in_message":[],"_revision_number":1},{"id":"2a282358506d72bebd5b962a106f3ed798b10776","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 15:57:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a3cfa526b4c249f9b8cd85c398a066c13d08e2b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 16:09:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/995ebfd5b6ca4017971c1200502e209e : SUCCESS in 1m 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c5eed4f5a794dc7bdbf4eb64fc107b5 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78682022d2dcf07c270356d23a27843578466469","revisions":{"78682022d2dcf07c270356d23a27843578466469":{"kind":"REWORK","_number":1,"created":"2022-01-11 18:09:57.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/58/824158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/824158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/824158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/824158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/824158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/824158/1"}}},"commit":{"parents":[{"commit":"bf1af8657f7ee101c213b6a60b9a60a1163ffd1e","subject":"Merge \"Dell PowerVault: Fix \"cinder manageable-list\"\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf1af8657f7ee101c213b6a60b9a60a1163ffd1e"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-07-09 15:08:34.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-01-11 18:09:57.000000000","tz":0},"subject":"Reject bad img formats for uploaded encrypted vols","message":"Reject bad img formats for uploaded encrypted vols\n\nCinder only supports uploading volumes of encrypted volume types as\nimages with disk format \u0027raw\u0027 and container format \u0027bare\u0027.  Screen\nfor this at the REST API layer when the request is made.\n\nChange-Id: Ibb77b8b1be6c35c5db3b07fdc4056afd51d48782\nCloses-bug: #1935688\n(cherry picked from commit de8b3b0b00421a01ed777880d2b5bf6fffc068d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78682022d2dcf07c270356d23a27843578466469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78682022d2dcf07c270356d23a27843578466469"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
