)]}'
{"id":"openstack%2Fcinder~825012","triplet_id":"openstack%2Fcinder~master~Iefd54ab86c048212e70b04bc09082bca0578ff07","project":"openstack/cinder","branch":"master","attention_set":{"28050":{"account":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"last_update":"2022-01-18 17:12:17.000000000","reason":"Miłosz Linkiewicz replied on the change"}},"removed_from_attention_set":{"34444":{"account":{"_account_id":34444,"name":"Miłosz Linkiewicz","email":"milosz.linkiewicz@intel.com","username":"Mionsz"},"last_update":"2022-01-18 17:12:17.000000000","reason":"Miłosz Linkiewicz replied on the change"}},"hashtags":[],"change_id":"Iefd54ab86c048212e70b04bc09082bca0578ff07","subject":"delete_nvmeof_target: detach controller before namespaces deletion","status":"NEW","created":"2022-01-17 20:24:13.000000000","updated":"2022-01-21 04:04:10.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f264094d34a8b2434c4ec5f886ff9824e0c03747","_number":825012,"virtual_id_number":825012,"owner":{"_account_id":34444,"name":"Miłosz Linkiewicz","email":"milosz.linkiewicz@intel.com","username":"Mionsz"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-18 12:03:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-17 20:24:22.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-17 20:26:57.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-01-17 20:29:21.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-17 20:31:05.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-01-17 21:14:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-17 21:33:40.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-01-17 21:44:18.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-17 21:54:05.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-01-17 22:52:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-18 00:01:57.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-18 02:15:58.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-18 03:19:00.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-01-18 04:26:06.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-01-18 08:24:12.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-18 12:07:47.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-18 13:14:32.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-01-18 13:21:45.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-01-18 13:28:01.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-18 14:58:03.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-18 15:07:25.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-01-18 15:49:14.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-01-18 16:05:35.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-18 17:12:17.000000000","updated_by":{"_account_id":34444,"name":"Miłosz Linkiewicz","email":"milosz.linkiewicz@intel.com","username":"Mionsz"},"reviewer":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"state":"REVIEWER"},{"updated":"2022-01-18 17:50:52.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-21 04:04:10.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"f08ab6d0343029e909a2aefc59bd75c1e0f31d16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34444,"name":"Miłosz Linkiewicz","email":"milosz.linkiewicz@intel.com","username":"Mionsz"},"date":"2022-01-17 20:24:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fa6686aee0d4615fed59a3bc290095bed14dbcc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-17 20:24:22.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/14931/","accounts_in_message":[],"_revision_number":1},{"id":"c51a0d466ac562afe66419a49f5d8d7abca8271e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-17 20:26:14.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/14931/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"07723b7ec61c57f0d1f7922b260650fef5d77b10","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-01-17 20:26:57.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"300c1adb23a9de71a3f983af945f86145bddf013","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34444,"name":"Miłosz Linkiewicz","email":"milosz.linkiewicz@intel.com","username":"Mionsz"},"date":"2022-01-17 20:28:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f6bf6814a80b0e5d42d170f14cd2f57071633510","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-17 20:29:21.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/12/825012/1/Cinder-tgtadm : ABORTED\n\n* SPDK http://13.74.249.42/refs/changes/12/825012/1/SPDK : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"eface09c76a29ec819f99f6fe2cd8912c79060e1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-17 20:31:05.000000000","message":"Patch Set 1:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/17_01_2022_20:24:21\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"a10650096c3c16a9e9d0768f0d699e24b3ed8dc1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-01-17 20:31:34.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"4fe661389348dc55e20c49e1d7162df5e87fd610","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-17 20:32:07.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/12/825012/2/Cinder-tgtadm : ABORTED\n\n* SPDK http://13.74.249.42/refs/changes/12/825012/2/SPDK : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"b45756a059a56b37620bbe197202e838689ff680","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-17 20:38:04.000000000","message":"Patch Set 2:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/17_01_2022_20:31:06\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"3131189c615bc601bc1482239bbf70d283ac2aa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34444,"name":"Miłosz Linkiewicz","email":"milosz.linkiewicz@intel.com","username":"Mionsz"},"date":"2022-01-17 20:38:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"acef00888d45571105e7a771545d0b0ae68b55db","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-17 20:38:36.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/12/825012/1/Cinder-tgtadm : ABORTED\n\n* SPDK http://13.74.249.42/refs/changes/12/825012/1/SPDK : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"8e762f810ea789414852f06019349e9335272961","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-17 20:39:00.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/12/825012/2/SPDK : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"d8133d17c88e28f3eac7fd76b2b192dd0e5f9db4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-01-17 20:42:48.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"16e815c82aa24cf7b39074551452cc8d518e926d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-17 20:52:02.000000000","message":"Patch Set 3:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/17_01_2022_20:44:47\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"361f7cb25e5c2a19c5619eebc33b7ad29db1358b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-17 20:52:34.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/12/825012/3/Cinder-tgtadm : ABORTED\n\n* SPDK http://13.74.249.42/refs/changes/12/825012/3/SPDK : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"60cdd3236148ac250129eda6eb87d90f6dc9040f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 21:14:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cbb09e92229e4ecdb258888786461e17 : SUCCESS in 15m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/814bc0c7a3f74f1fa49804fc2071a09f : FAILURE in 29m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e1b30589b9f3864e7c62caf08abf2e5478be66ce","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-01-17 21:33:40.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/12/825012/3/check/iscsi-driver-alletra-6k/98a1cc8/98a1cc8.html : FAILURE in 50m 22s","accounts_in_message":[],"_revision_number":3},{"id":"0edf7182e6300e8574cf4f261e990c4cde7f2c5b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-17 21:44:18.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/12/825012/3/check/tempest-dsvm-full-focal-py3/4849111/ : FAILURE in 58m 31s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/12/825012/3/check/devstack-plugin-ceph-dsvm-focal-py3/3b670e6/ : FAILURE in 56m 25s","accounts_in_message":[],"_revision_number":3},{"id":"ef1537fb36856baf25bac9393249181509c0fb80","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-17 21:44:29.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/12/825012/3/Cinder-tgtadm : ABORTED\n\n* SPDK http://13.74.249.42/refs/changes/12/825012/3/SPDK : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"b8202019fe04de5ba8eceeeb492fc2a63d537de4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-17 21:54:05.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825012/3/iscsi/ : SUCCESS in 1h 14m 58s\n- cinder-smb http://cloudbase-ci.com/cinder/825012/3/smb/ : SUCCESS in 1h 06m 14s","accounts_in_message":[],"_revision_number":3},{"id":"26faf9bfd2bcb2f1652ab254096075e3abf8f3a6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-17 22:24:58.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/825012/825012/3/check-not-vote/ext-cinder-zuul/cb2abc8 : FAILURE in 44m 31s","accounts_in_message":[],"_revision_number":3},{"id":"b0e4a8064e6c9dcc3f864baa60b68592f5ca2840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 22:52:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47abbb502bf041338fb4223456c62c9a : FAILURE in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e681792b78a488fa7760cf42828fb43 : SUCCESS in 10m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0845d2f8ed6542a8b61f30df19952ced : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b1c522c0d5c4c55917c287025605d02 : SUCCESS in 14m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/6c4f4c8cb7f24504986dc229b9f849fe : SUCCESS in 45m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/666e05240a9948cbbf1067263d2f5ac8 : SUCCESS in 1h 06m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/371d33ca43f14dd9ac71723e7d7deca6 : SUCCESS in 55m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/80d29422c9784286a68470b9cd0be007 : SUCCESS in 13m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9085d70ecd694cb498d967b314e3bd0a : FAILURE in 5m 28s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/22c8508a8e944c6cb87a4be7c5dc450a : SUCCESS in 9m 22s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7db64fba5c774dcc98426e66bc662549 : SUCCESS in 7m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/639f4efd20a34ff2bfba1244c376802f : SUCCESS in 6m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/47cd989a10b54c71a692d1d2aa491550 : SUCCESS in 39m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f7072d0af8c541a7a0f25ed057c707a2 : SUCCESS in 7m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c3bef74fe8a24f6181d26bb0fbc48cf2 : SUCCESS in 1h 22m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3232b710506d4b06bfbb9f846f7d5c79 : SUCCESS in 1h 33m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a0c8032cbebd47c2bee0fe91a34a2ec9 : SUCCESS in 1h 58m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bbb4d58ed1604c90911b4921c45282af : SUCCESS in 48m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7e508afc483147b2850fd01d8e284330 : SUCCESS in 1h 27m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d09f207b0e6646eeaf8944bc48d52765 : SUCCESS in 38m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/eaec1b85383d4bc4b2713fecebc75142 : SUCCESS in 1h 13m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/93970c5f0eaa4c1997126f949fb5eda7 : SUCCESS in 2h 09m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/734cd2ba78b248b9a87ccf0d3ec8a4dc : SUCCESS in 54m 48s\n\nWarning:\n  Comments left for invalid file cinder/volume/manager.py","accounts_in_message":[],"_revision_number":3},{"id":"f56b228e22507776e29d8ede879b4abaf2bd3f28","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-17 23:21:14.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/12/825012/3/Cinder-tgtadm : FAILURE\n\n* SPDK http://13.74.249.42/refs/changes/12/825012/3/SPDK : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"3c32220b87a8f459609499fdf7119b8c87364f61","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-17 23:44:25.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/14936/","accounts_in_message":[],"_revision_number":2},{"id":"122331e0e7691125bfc82358e9f76058477d76b0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-17 23:46:04.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/14936/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"6621826df7b997db70dd8340dab55d5f843914b2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-01-18 00:01:57.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/621073e464a842b9822a84604833efc3 : FAILURE in 1h 11m 46s","accounts_in_message":[],"_revision_number":3},{"id":"6bda9c832050a5910dacbd407a6fffaab86af356","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-18 01:43:20.000000000","message":"Patch Set 3:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":3},{"id":"048410161444bc2b07434413799f711b10144a34","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-01-18 02:15:58.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/12/825012/3/check/DellEMC_XtremIO_FC/351ff32/DellEMC_XtremIO_FC/489 : SUCCESS in 1h 02m 46s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/12/825012/3/check/DellEMC_XtremIO_iSCSI/501ac78/DellEMC_XtremIO_iSCSI/500 : SUCCESS in 1h 34m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"05791f031b9aa8dcd45215cfae8e6422082e4369","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-18 03:19:00.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/12/825012/3/check/DellEMC_Unity_FC/0ea8aba/DellEMC_Unity_FC/412 : FAILURE in 1h 18m 03s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/12/825012/3/check/DellEMC_Unity_iSCSI/376d1f2/DellEMC_Unity_iSCSI/400 : SUCCESS in 1h 32m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"c573affd089c0d34466f474bf4e20ac408331397","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-18 03:19:38.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/12/825012/3/Cinder-tgtadm : ABORTED\n\n* SPDK http://13.74.249.42/refs/changes/12/825012/3/SPDK : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"7c10fae24f0416a7a11aae5c8438e5f02e46b516","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-18 04:08:30.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/12/825012/3/Cinder-tgtadm : FAILURE\n\n* SPDK http://13.74.249.42/refs/changes/12/825012/3/SPDK : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"187c9986e8c9cfbf992cc7ed804d3131b0425690","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-01-18 04:26:06.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/12/825012/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/6b074be/ : TIMED_OUT in 3h 05m 54s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/12/825012/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6f931d4/ : SUCCESS in 1h 02m 43s","accounts_in_message":[],"_revision_number":3},{"id":"34536f0ded7fc46e680e4dd71b9fcab37f2d3775","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-01-18 08:24:12.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/12/825012/3/ : FAILURE in 1h 12m 09s","accounts_in_message":[],"_revision_number":3},{"id":"0f874793daaa36fd7aec19b42b1393d5029de393","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34444,"name":"Miłosz Linkiewicz","email":"milosz.linkiewicz@intel.com","username":"Mionsz"},"date":"2022-01-18 09:23:55.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f78ea7c41b57f0277952ab22788bcbaa86b5042a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 10:02:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f9b76d71c4a4a6e9059704d0e433a7e : SUCCESS in 21m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b93d40ee046e43718e765170e72d059e : SUCCESS in 19m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"75339d9a03bdbad37a22c7bae4da4398a0e59e44","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-01-18 10:23:58.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/12/825012/4/check/iscsi-driver-alletra-6k/0823d74/0823d74.html : FAILURE in 50m 07s","accounts_in_message":[],"_revision_number":4},{"id":"c70eaf2849e1a500b6d46ba73a30929e32c306fd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-18 10:24:03.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825012/4/iscsi/ : SUCCESS in 59m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/825012/4/smb/ : SUCCESS in 52m 34s","accounts_in_message":[],"_revision_number":4},{"id":"bdfd008d1c3a6dc6f05945837ce8815eb34b58d9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-18 10:41:35.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/12/825012/4/Cinder-tgtadm : FAILURE\n\n* SPDK http://13.74.249.42/refs/changes/12/825012/4/SPDK : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"5765e0027434c67b5a2c1ee41d557476d8b1db45","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-18 10:47:55.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/12/825012/4/check/tempest-dsvm-full-focal-py3/ebfcfd8/ : FAILURE in 1h 09m 06s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/12/825012/4/check/devstack-plugin-ceph-dsvm-focal-py3/5e388d9/ : FAILURE in 1h 09m 23s","accounts_in_message":[],"_revision_number":4},{"id":"5cb873eb8589f8fe74e49a0765b22adb95a88389","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-18 11:44:42.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/12/825012/4/Cinder-tgtadm : FAILURE\n\n* SPDK http://13.74.249.42/refs/changes/12/825012/4/SPDK : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"05d109a3af42fccb55b1f2ecec91f568cfc5daa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 12:03:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b95d1b5b102748cd9667a20a95eb3a1f : SUCCESS in 8m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f868b1ed9a6c484580bd01ae7b388a88 : SUCCESS in 11m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c080dab16465419cbf75ad89a42d02e8 : SUCCESS in 11m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ed570f032924e65814e46c4f0feee93 : SUCCESS in 13m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/3a286ab1d9fa436db0a2817260844b6a : SUCCESS in 56m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3428e3e036c54afb8107f10ae7ba9c67 : SUCCESS in 1h 08m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa79d5998e39445ba36fa3ababe55a35 : SUCCESS in 47m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4019de2e5d8b40799d241157a3629938 : SUCCESS in 15m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/eee82fa0222140f08c0c2111e9846046 : FAILURE in 8m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a70e3a6774e34fa18f5191db06380c99 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e13fe522feb3447ab7f27ef50c4dd4cd : SUCCESS in 8m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7b183523b8254877a4e55a7075a8355d : SUCCESS in 8m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0f7a7e432573453c899e909dc82836ad : SUCCESS in 40m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/346681025494400c99a92c008359c3ec : SUCCESS in 7m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9fc34cb4e046479b803487c166010ad2 : SUCCESS in 1h 21m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/64ee55925d244376b178435b4acd33c6 : SUCCESS in 1h 42m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bdf776ebfc3049e39fdb2cac5dd38c8b : SUCCESS in 1h 48m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ddd5461952c34414a6a3f57f3cd4a803 : SUCCESS in 1h 06m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/782e4694f0cf4328b8301b9c2908a25a : SUCCESS in 1h 20m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bd0d1fe8409042019728006493690281 : SUCCESS in 48m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3ba390db0c7e4a7da3207d89e8a957eb : SUCCESS in 1h 15m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bc33b5aa0b0e47379d10ab99f5570304 : SUCCESS in 2h 22m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dea6d12d0b3e400abf9b69ad211dde5c : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":4},{"id":"cd2f38f8ff853e2529215c9788696d9e09399873","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-18 12:07:47.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/12/825012/4/check/DellEMC_PowerFlex/891bd0a/DellEMC_PowerFlex/273 : FAILURE in 1m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"bdcb34a41f7f437bb340f45e0db41ad5f1491697","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-01-18 12:20:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2f2b042eefd0413fb2a994dc2307ba77 : FAILURE in 1h 23m 23s","accounts_in_message":[],"_revision_number":4},{"id":"5c797b3b293c5b35c370f5815505e1431c24f9bf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-18 12:26:56.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/12/825012/4/check/DellEMC_Unity_FC/3bea6d6/DellEMC_Unity_FC/418 : FAILURE in 1h 21m 37s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/12/825012/4/check/DellEMC_Unity_iSCSI/dffeced/DellEMC_Unity_iSCSI/403 : FAILURE in 5m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"e61defe0083867dca53e41c6e3942c0e5b268c3b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-18 12:27:33.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/12/825012/4/SPDK : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"a28dee439c176591c42aeff365218e014c24e24f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-01-18 12:29:53.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/12/825012/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2e003fd/ : TIMED_OUT in 3h 03m 19s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/12/825012/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/fddc74c/ : TIMED_OUT in 3h 03m 32s","accounts_in_message":[],"_revision_number":4},{"id":"0b450bc669e5943a90bf6e1d19f0cceef357fc43","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-01-18 12:34:30.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/12/825012/4/check/DellEMC_XtremIO_FC/1578805/DellEMC_XtremIO_FC/492 : SUCCESS in 57m 06s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/12/825012/4/check/DellEMC_XtremIO_iSCSI/7dfd85a/DellEMC_XtremIO_iSCSI/503 : SUCCESS in 1h 35m 34s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"903038954513c811569c19656e8f11b6a8b028fe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-18 12:34:49.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/12/825012/4/SPDK : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"bccdae3b389bb3e5f1aeb0c62aae52efee7da561","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-18 12:39:21.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/825012/825012/4/check-not-vote/ext-cinder-zuul/a842d28 : FAILURE in 44m 59s","accounts_in_message":[],"_revision_number":4},{"id":"fb2616d4b571445b0176ba339bf7e80a38a6c11e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-18 12:40:37.000000000","message":"Patch Set 4:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/18_01_2022_12:33:25\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"03579665f0c13463d4547f5c91d10ae8dbfd5521","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-18 12:40:56.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/12/825012/4/SPDK : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"9d33f50231c1a082de78fa2c728b435c281d4d23","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-01-18 13:00:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/12/825012/4/check/dsvm-tempest-cisco-zonemanager-job/5e214d4/ : SUCCESS in 1h 03m 13s","accounts_in_message":[],"_revision_number":4},{"id":"03e105d44cf37817ba226e2d4372317c1fc38c48","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-01-18 13:14:32.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/12/825012/4/upstream-solidfire-check/cinder-solidfire-iSCSI/5d55459/ : SUCCESS in 1h 06m 30s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"a312c7008fd57efe04c06fa101fd35f4864461d9","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-01-18 13:21:45.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-12-825012-4/1642507381 : SUCCESS in 35m 03s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"03e32b2325753b779dbb30ea73ef3f74a2553d23","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-18 13:28:01.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/12/825012/4/upstream-check/cinder-cDOT-iSCSI/e318c61/ : SUCCESS in 42m 27s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/12/825012/4/upstream-check/cinder-cDOT-NFS/789f761/ : SUCCESS in 42m 32s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/12/825012/4/upstream-check/cinder-cDOT-FCP/ccb8f18/ : SUCCESS in 1h 20m 05s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"6eecfe19f790c55a83bfc22e7c79c9fb44d97f63","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-18 14:18:31.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/12/825012/4/Cinder-tgtadm : FAILURE\n\n* SPDK http://13.74.249.42/refs/changes/12/825012/4/SPDK : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"45a27cee5d1119b84e53ddd406350b6d95c865da","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-01-18 14:58:03.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/12/825012/4/check/DellEMC_PowerMAX_FC/eddd4b6/DellEMC_PowerMAX_FC/188 : FAILURE in 1h 27m 09s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/12/825012/4/check/DellEMC_PowerMAX_iSCSI/2d0505e/DellEMC_PowerMAX_iSCSI/201 : FAILURE in 1h 14m 55s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"c779f43a5908edee30d3b0737ba9f586664eb030","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-01-18 15:07:25.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/12/825012/4/check/DellEMC_PowerStore_iSCSI/b4f60c9/DellEMC_PowerStore_iSCSI/243 : SUCCESS in 1h 26m 53s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/12/825012/4/check/DellEMC_PowerStore_FC/0f20cf8/DellEMC_PowerStore_FC/247 : SUCCESS in 1h 09m 37s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"fb9e26151e257960c4bfdc9c814540888d2e3c3c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-18 15:07:40.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/12/825012/4/SPDK : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"10dfeead9e686a24c9c61b7ae15f77ddb9218558","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-01-18 15:49:14.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/12/825012/4/check/DellEMC_SC_FC/ae7691f/DellEMC_SC_FC/211 : SUCCESS in 48m 45s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/12/825012/4/check/DellEMC_SC_iSCSI/4abade8/DellEMC_SC_iSCSI/208 : SUCCESS in 1h 53m 31s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"d19077a4fbc43797db78a0e205e18b91482352e6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-18 15:49:38.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/12/825012/4/SPDK : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"df5c3e6bf43a3e3265e7d419ba0e6a474a542e1a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-18 16:05:35.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-12-825012-3/ns5_nfs : SUCCESS in 5h 3m 45s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-12-825012-3/ns5_iscsi : SUCCESS in 5h 7m 19s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"421c3774f55719cdf89993f85cd414201bfce955","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-18 16:06:04.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/12/825012/4/SPDK : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"186fcdb3adcba58224269bd05f8c64d535945ee7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-18 16:28:47.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-12-825012-2/ns5_nfs : SUCCESS in 5h 25m 56s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-12-825012-2/ns5_iscsi : SUCCESS in 5h 17m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"5b7ef02410590b69979dee2a564202ead5d928d7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-18 16:29:37.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/12/825012/3/SPDK : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"069e2da87c19cb7823ab4cbbf279d51be61d2bd8","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-01-18 17:50:52.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/12/825012/4/check/DellEMC_VNX_iSCSI/c68af0a/DellEMC_VNX_iSCSI/208 : FAILURE in 2h 53m 00s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/12/825012/4/check/DellEMC_VNX_FC/1bab97f/DellEMC_VNX_FC/218 : FAILURE in 2h 11m 08s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"efbce29beef732d9a25e4588f3441cc02c39d164","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-18 17:51:30.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/12/825012/2/Cinder-tgtadm : ABORTED\n\n* SPDK http://13.74.249.42/refs/changes/12/825012/2/SPDK : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"561f8e74789ec93741cfa21d3d8c803e63387d9d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-18 19:15:55.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-12-825012-4/ns5_nfs : SUCCESS in 5h 5m 39s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-12-825012-4/ns5_iscsi : SUCCESS in 5h 2m 12s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"84db02f4a1f7534ad11a76f2fa339231f48ab919","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-18 19:16:42.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/12/825012/4/Cinder-tgtadm : ABORTED\n\n* SPDK http://13.74.249.42/refs/changes/12/825012/4/SPDK : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"481d3b143b98862a7a6ff3f51de2170580ac676c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-18 21:53:57.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/12/825012/4/Cinder-tgtadm : FAILURE\n\n* SPDK http://13.74.249.42/refs/changes/12/825012/4/SPDK : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"d0f4f365293408da36a0a5079f021992faa4a0a3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-19 01:53:50.000000000","message":"Patch Set 4:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":4},{"id":"ca8a4722cfaa79f08b434443d51bf8aca3878785","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-19 02:14:30.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-12-825012-1/ns5_nfs : SUCCESS in 5h 18m 10s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-12-825012-1/ns5_iscsi : FAILURE in 17h 29m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"7ba2d5b9e2bc4072773bcb0740399e640af866cf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-19 05:31:05.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/12/825012/1/Cinder-tgtadm : FAILURE\n\n* SPDK http://13.74.249.42/refs/changes/12/825012/1/SPDK : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"b0f2a6b7c9d8321357010535a02a863b20fb6344","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-01-19 18:40:18.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/12/825012/4/ : SUCCESS in 1h 50m 24s","accounts_in_message":[],"_revision_number":4},{"id":"f264094d34a8b2434c4ec5f886ff9824e0c03747","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-01-21 04:04:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/12/825012/4/20220121101112/NEC-FC : SUCCESS in 2h 43m 55s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/12/825012/4/20220121101112/NEC-ISCSI : SUCCESS in 2h 52m 55s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bbcfbb0a26b9a62c06f742caa971c302d613d6cf","revisions":{"69ecb0782f82835caae5eb644cd8376964b4accf":{"kind":"REWORK","_number":1,"created":"2022-01-17 20:24:13.000000000","uploader":{"_account_id":34444,"name":"Miłosz Linkiewicz","email":"milosz.linkiewicz@intel.com","username":"Mionsz"},"ref":"refs/changes/12/825012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/12/825012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/825012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/825012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/825012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/12/825012/1"}}},"commit":{"parents":[{"commit":"d4fe88694d7bdd2379256913af2b6e35bf1b2ed0","subject":"delete_nvmeof_target: remove all namespaces before subsystem delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d4fe88694d7bdd2379256913af2b6e35bf1b2ed0"}]}],"author":{"name":"mionsz","email":"milosz.linkiewicz@intel.com","date":"2022-01-17 20:23:31.000000000","tz":0},"committer":{"name":"mionsz","email":"milosz.linkiewicz@intel.com","date":"2022-01-17 20:23:31.000000000","tz":0},"subject":"delete_nvmeof_target: detach controller before namespaces deletion","message":"delete_nvmeof_target: detach controller before namespaces deletion\n\nDetach controller needs to be called before namespaces removal. The RPC call bdev_nvme_detach_controller removes all bdevs from the subsystem and then removes the controller itself. When RPC is called with all mandatory \u0027trid\u0027 subfields and ctrl name, the \"ctrlrs\" structure is removed from the controller, when no more ctrlrs are present, the controller itself is removed.\n\nSigned-off-by: mionsz \u003cmilosz.linkiewicz@intel.com\u003e\nChange-Id: Iefd54ab86c048212e70b04bc09082bca0578ff07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/69ecb0782f82835caae5eb644cd8376964b4accf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/69ecb0782f82835caae5eb644cd8376964b4accf"}]},"branch":"refs/heads/master"},"56ea135577f0bce1a9bdfa9dc457ae4eb8e04f30":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-17 20:28:08.000000000","uploader":{"_account_id":34444,"name":"Miłosz Linkiewicz","email":"milosz.linkiewicz@intel.com","username":"Mionsz"},"ref":"refs/changes/12/825012/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/12/825012/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/825012/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/825012/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/825012/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/12/825012/2"}}},"commit":{"parents":[{"commit":"895e46477c1cf760f57ca282e2fe81690ced71cd","subject":"requests.packages.urllib3.disable_warnings: Module patch check for attributes.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/895e46477c1cf760f57ca282e2fe81690ced71cd"}]}],"author":{"name":"mionsz","email":"milosz.linkiewicz@intel.com","date":"2022-01-17 20:23:31.000000000","tz":0},"committer":{"name":"Miłosz Linkiewicz","email":"milosz.linkiewicz@intel.com","date":"2022-01-17 20:28:08.000000000","tz":0},"subject":"delete_nvmeof_target: detach controller before namespaces deletion","message":"delete_nvmeof_target: detach controller before namespaces deletion\n\nDetach controller needs to be called before namespaces removal. The RPC call bdev_nvme_detach_controller removes all bdevs from the subsystem and then removes the controller itself. When RPC is called with all mandatory \u0027trid\u0027 subfields and ctrl name, the \"ctrlrs\" structure is removed from the controller, when no more ctrlrs are present, the controller itself is removed.\n\nSigned-off-by: mionsz \u003cmilosz.linkiewicz@intel.com\u003e\nChange-Id: Iefd54ab86c048212e70b04bc09082bca0578ff07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56ea135577f0bce1a9bdfa9dc457ae4eb8e04f30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56ea135577f0bce1a9bdfa9dc457ae4eb8e04f30"}]},"branch":"refs/heads/master","description":"Rebase"},"fec5801a555e6ddba4a0c6a946c5937ebacbb650":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-17 20:38:25.000000000","uploader":{"_account_id":34444,"name":"Miłosz Linkiewicz","email":"milosz.linkiewicz@intel.com","username":"Mionsz"},"ref":"refs/changes/12/825012/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/12/825012/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/825012/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/825012/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/825012/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/12/825012/3"}}},"commit":{"parents":[{"commit":"d367f4f30ffe7e5d1472ecc34bdaa8143c3a9a9e","subject":"requests.packages.urllib3.disable_warnings: Module patch check for attributes.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d367f4f30ffe7e5d1472ecc34bdaa8143c3a9a9e"}]}],"author":{"name":"mionsz","email":"milosz.linkiewicz@intel.com","date":"2022-01-17 20:23:31.000000000","tz":0},"committer":{"name":"Miłosz Linkiewicz","email":"milosz.linkiewicz@intel.com","date":"2022-01-17 20:38:25.000000000","tz":0},"subject":"delete_nvmeof_target: detach controller before namespaces deletion","message":"delete_nvmeof_target: detach controller before namespaces deletion\n\nDetach controller needs to be called before namespaces removal. The RPC call bdev_nvme_detach_controller removes all bdevs from the subsystem and then removes the controller itself. When RPC is called with all mandatory \u0027trid\u0027 subfields and ctrl name, the \"ctrlrs\" structure is removed from the controller, when no more ctrlrs are present, the controller itself is removed.\n\nSigned-off-by: mionsz \u003cmilosz.linkiewicz@intel.com\u003e\nChange-Id: Iefd54ab86c048212e70b04bc09082bca0578ff07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fec5801a555e6ddba4a0c6a946c5937ebacbb650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fec5801a555e6ddba4a0c6a946c5937ebacbb650"}]},"branch":"refs/heads/master","description":"Rebase"},"bbcfbb0a26b9a62c06f742caa971c302d613d6cf":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-01-18 09:23:55.000000000","uploader":{"_account_id":34444,"name":"Miłosz Linkiewicz","email":"milosz.linkiewicz@intel.com","username":"Mionsz"},"ref":"refs/changes/12/825012/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/12/825012/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/825012/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/825012/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/825012/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/12/825012/4"}}},"commit":{"parents":[{"commit":"7397cc36f41f9d5d59d4ea6ce5e82a6a20969174","subject":"requests.packages.urllib3.disable_warnings: Module patch check for attributes.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7397cc36f41f9d5d59d4ea6ce5e82a6a20969174"}]}],"author":{"name":"mionsz","email":"milosz.linkiewicz@intel.com","date":"2022-01-17 20:23:31.000000000","tz":0},"committer":{"name":"Miłosz Linkiewicz","email":"milosz.linkiewicz@intel.com","date":"2022-01-18 09:23:55.000000000","tz":0},"subject":"delete_nvmeof_target: detach controller before namespaces deletion","message":"delete_nvmeof_target: detach controller before namespaces deletion\n\nDetach controller needs to be called before namespaces removal. The RPC call bdev_nvme_detach_controller removes all bdevs from the subsystem and then removes the controller itself. When RPC is called with all mandatory \u0027trid\u0027 subfields and ctrl name, the \"ctrlrs\" structure is removed from the controller, when no more ctrlrs are present, the controller itself is removed.\n\nSigned-off-by: mionsz \u003cmilosz.linkiewicz@intel.com\u003e\nChange-Id: Iefd54ab86c048212e70b04bc09082bca0578ff07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bbcfbb0a26b9a62c06f742caa971c302d613d6cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bbcfbb0a26b9a62c06f742caa971c302d613d6cf"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
