)]}'
{"id":"openstack%2Fcinder~825100","triplet_id":"openstack%2Fcinder~stable%2Fvictoria~Ie48cf55deacd08e7716201dac00ede8d57e6632f","project":"openstack/cinder","branch":"stable/victoria","topic":"victoria-detach-vol-status","attention_set":{},"removed_from_attention_set":{"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-02-09 16:37:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie48cf55deacd08e7716201dac00ede8d57e6632f","subject":"Fix detach notification","status":"MERGED","created":"2022-01-18 11:54:40.000000000","updated":"2022-02-09 16:37:44.000000000","submitted":"2022-02-09 16:37:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825100-victoria-detach-vol-status","meta_rev_id":"4d1506833dbeeee893edc8b6d6314ceba7615002","_number":825100,"virtual_id_number":825100,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-09 16:37:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-02-06 04:50:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-01-20 06:39:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-09 14:46:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-09 14:46:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-18 12:47:41.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-01-18 13:03:38.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-01-18 14:14:27.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-01-18 14:20:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-18 15:20:34.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-18 15:29:11.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-18 16:02:51.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-01-18 17:03:38.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-18 19:40:02.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-18 20:09:56.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-01-20 06:39:38.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2022-01-21 10:06:48.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-06 04:50:15.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-02-09 14:46:19.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"af5adad6fd768c46e3b72975bc9d21a92d1a6646","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-01-18 11:54:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12669642948e0e26c02606fd98ff322c0f3b698f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-01-18 12:47:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/00/825100/1/check/iscsi-driver-alletra-6k/e1ffaa0/e1ffaa0.html : FAILURE in 51m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1ad861e9e0056220203a474e50a3365df2bbaaaf","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-18 13:03:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825100/1/iscsi/ : SUCCESS in 1h 08m 12s\n- cinder-smb http://cloudbase-ci.com/cinder/825100/1/smb/ : SUCCESS in 57m 29s","accounts_in_message":[],"_revision_number":1},{"id":"8bc95022e554c77dc0d5c02408c8a97f2d356cf7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-01-18 14:14:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/00/825100/1/check/dsvm-tempest-cisco-zonemanager-job/d087fc2/ : FAILURE in 1h 08m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6a1dca9d3d8eb9f8cad7e3299ae92f70c6fdd49c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 14:20:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7903530e3ecc4f819969136f27e52484 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/368b64fb7ac3483b8fad4c1af3c95633 : SUCCESS in 9m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/225c50d6d20b4da499e3b4febf0fda0c : SUCCESS in 9m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dc69534d7864725ac21584bfc1a7570 : SUCCESS in 16m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/d38c7bd65b474a72aad2d58d048f7e2f : SUCCESS in 1h 06m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/10bf260408834d7e8df52e96ad7719ab : SUCCESS in 1h 07m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c741817691ab4a51a4da2088398fcd7e : SUCCESS in 39m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39e3418792f04ce99140b45bfd01ad71 : SUCCESS in 6m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9ce2d5ebf84f4637a1584673d8f51000 : SUCCESS in 13m 12s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1e73380dde824f12be48a55f948a0116 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/683ca5936aa04500afb712aa5e98fb43 : SUCCESS in 7m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d729abbb448147abb70cf26c38bbb871 : SUCCESS in 7m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3b1c363cfec34c67874a2ef9fddeb098 : SUCCESS in 39m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4c1427c84999447597833c10b8bbb5c8 : FAILURE in 8m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d483eb7c15cf439bad9c2b9163ff20dd : SUCCESS in 1h 15m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8562fb0bd7cf473cacd38bdfbfbded5e : FAILURE in 1h 52m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c372b59af50348d6899b41d94bd23ad0 : SUCCESS in 1h 35m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b4d4506e6260497bb896d53f43f1a3d0 : SUCCESS in 1h 08m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4312a863b674420b91a411c1f17500ce : SUCCESS in 1h 27m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9dba0c13cadd464480ee1a27fd7ccb94 : SUCCESS in 1h 12m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/43f1bfd9818e409992ac336d492f40eb : SUCCESS in 2h 21m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e3b9fc885bb41f1a671e5682fd668fc : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":1},{"id":"fa3396274379a171be1d767d036b575f2db913d7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-18 15:20:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/00/825100/1/upstream-check/cinder-cDOT-iSCSI/a415c2b/ : FAILURE in 35m 01s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/00/825100/1/upstream-check/cinder-cDOT-NFS/2135061/ : FAILURE in 35m 44s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/00/825100/1/upstream-check/cinder-cDOT-FCP/7cd6ab2/ : FAILURE in 55m 35s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"dedebd9330598031dcd3accd6c7f8474fca7086c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-18 15:29:11.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/825100/1/check/tempest-dsvm-full-focal-py3/6428634/ : SUCCESS in 2h 28m 42s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/825100/1/check/devstack-plugin-ceph-dsvm-focal-py3/3726c0f/ : FAILURE in 1h 16m 44s","accounts_in_message":[],"_revision_number":1},{"id":"cdf4bb955891d111074fe1bcc59e5992125ba69e","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-01-18 16:02:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/00/825100/1/upstream-solidfire-check/cinder-solidfire-iSCSI/b1d8009/ : FAILURE in 1h 37m 36s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"1d717b9a85b965a0547cfe02459d3059dedb4237","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-01-18 17:03:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/00/825100/1/ : SUCCESS in 1h 31m 02s","accounts_in_message":[],"_revision_number":1},{"id":"9ac05b53d8eb6956a299e2cb514bc1585471244f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-18 19:40:02.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-00-825100-1/ns5_nfs : SUCCESS in 5h 27m 32s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-00-825100-1/ns5_iscsi : SUCCESS in 5h 3m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"5a193bc2eaf0790015ca8f276cc94831355605e7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-01-18 20:09:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/00/825100/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/abeaffb/ : TIMED_OUT in 3h 02m 24s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/00/825100/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/612fdc3/ : SUCCESS in 1h 06m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0bee20376226a1b0de710e6f8e090ff103ac94b3","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-01-20 06:39:38.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1415465e87e901e3e0fd957d4afcc7dc761cb146","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-01-21 10:06:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/825100/1/20220121130414/NEC-FC : SUCCESS in 2h 52m 24s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/825100/1/20220121130414/NEC-ISCSI : SUCCESS in 6h 02m 32s","accounts_in_message":[],"_revision_number":1},{"id":"09e00dd474707d790b57e39a9e4432c1c67601f5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-06 04:50:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0aa8346735b9c193002f79582246582e8ebef6ef","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-02-09 14:46:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b74ed94bcf8bece602b98d79caf50bf1409c7adc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 15:00:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2cc532d79941170383b651bd72ddae1cabc459a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 16:37:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff6dbd24aa994ae39a97ab761999bf69 : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5ef9311588d46d3bf80d27d32685a1f : SUCCESS in 9m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab284b8e653f402386c82dccafaf1d09 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e3525bf72044de8024e1e1c2f9c547 : SUCCESS in 14m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/d873509c46144678b1af799c67588d04 : SUCCESS in 52m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87fc725d7b7f43be924d5443ca6e219e : SUCCESS in 1h 01m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54430eea5fcd4c389990982a5e014459 : SUCCESS in 55m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/482d12331e854aae8894ecd0d8e22d17 : SUCCESS in 7m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ee35899f44104f41b063fc429905d8d6 : SUCCESS in 1h 14m 26s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/be7f1d4dd0604e02ac38d50ab7c38b25 : SUCCESS in 1h 03m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e007389992b4cc3a17315a54960312c : SUCCESS in 1h 04m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4d1506833dbeeee893edc8b6d6314ceba7615002","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 16:37:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9d3cc966cc4856b9fcc8625ce3ce86061df1d87","revisions":{"c9d3cc966cc4856b9fcc8625ce3ce86061df1d87":{"kind":"REWORK","_number":1,"created":"2022-01-18 11:54:40.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/00/825100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/825100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/825100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/825100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/825100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/825100/1"}}},"commit":{"parents":[{"commit":"7d6ca2feb920823694307e9da63a92ce76734de9","subject":"Merge \"3PAR: Allow iSCSI driver to be enabled for Primera 4.2 onwards.\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d6ca2feb920823694307e9da63a92ce76734de9"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-07-02 13:11:53.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-01-18 09:35:33.000000000","tz":60},"subject":"Fix detach notification","message":"Fix detach notification\n\nOur current `attachment_delete` methods in the volume API and the\nmanager are using DB methods directly, which makes the OVOs present in\nthose methods get out of sync with the latest data, which leads to\nnotifications having the wrong data when we send them on volume detach.\n\nThis patch replaces DB method calls with OVO calls and moves the\nnotification call to the end of the method, where we have the final\nstatus on the volume.\n\nIt also adds the missing detach.start notification when deleting an\nattachment in the reserved state.\n\nCloses-Bug: #1916980\nCloses-Bug: #1935011\nChange-Id: Ie48cf55deacd08e7716201dac00ede8d57e6632f\n(cherry picked from commit 68d49445778ef486e3ff656929405ab270a5a65d)\nConflicts:\n\tcinder/volume/api.py\nChanges:\n        cinder/volume/manager.py\n(cherry picked from commit c0197c6f1ad13e3a63615ceb667a53d3bbca2d77)\n(cherry picked from commit ed06fc7452699d76c22b288e1cc8e8bc2afa1849)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9d3cc966cc4856b9fcc8625ce3ce86061df1d87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9d3cc966cc4856b9fcc8625ce3ce86061df1d87"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
