)]}'
{"id":"openstack%2Fcinder~825105","triplet_id":"openstack%2Fcinder~stable%2Fvictoria~I6576832b2c2722ab749cfe70bbc2058ead816c36","project":"openstack/cinder","branch":"stable/victoria","topic":"victoria-detach-vol-status","attention_set":{},"removed_from_attention_set":{"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-02-10 17:00:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6576832b2c2722ab749cfe70bbc2058ead816c36","subject":"Expose volume_attachments in Volume OVO","status":"MERGED","created":"2022-01-18 11:54:40.000000000","updated":"2022-02-10 17:12:00.000000000","submitted":"2022-02-10 17:00:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825105-victoria-detach-vol-status","meta_rev_id":"949a292ebfee6affb7aa6e75af37c7aac3c99cf8","_number":825105,"virtual_id_number":825105,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-10 17:00:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-02-10 13:30:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-10 14:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-10 14:46:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-02-10 13:30:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-18 14:12:22.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-01-18 14:30:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-18 14:34:43.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-01-18 15:31:04.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-18 17:00:12.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-01-18 20:56:07.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-18 21:57:53.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-18 22:14:25.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-01-22 10:17:14.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-31 17:45:58.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2022-02-06 04:56:47.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2022-02-09 14:53:13.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"CC"},{"updated":"2022-02-10 13:30:37.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-02-10 14:46:17.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"b9a8542650f58938e5d526363e137688f2d3ac5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-01-18 11:54:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"852536d0b76338fa1ea0fbfed610eed22c2f5946","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-18 14:12:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825105/1/iscsi/ : SUCCESS in 1h 01m 47s\n- cinder-smb http://cloudbase-ci.com/cinder/825105/1/smb/ : SUCCESS in 1h 15m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0d441f9547937fa9e4cc383600bbb50d2b12d347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 14:30:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef6bacd6e93d4ec4b70b3997b598d8f9 : SUCCESS in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f648d685358425e8c5e0acaab3a8b81 : SUCCESS in 10m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be1af5c13e8b4915bf95dc69fc9685c0 : SUCCESS in 10m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de4816b5378a451ab743049a40709884 : SUCCESS in 16m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/9efd7ed05765472f8aabea5bf9e91382 : SUCCESS in 1h 09m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b01b1715ab5f47be889b8a5398e27e10 : SUCCESS in 1h 14m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e82641dcc3d4c7b9c41b623c32ca9ba : SUCCESS in 54m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5cc74b58e8144a1681cd84c75f3d7bc7 : SUCCESS in 13m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/69b562f8bbe94022ae90c5d96d24494d : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e02baef8895c4b17a8ea162d573f5c6d : SUCCESS in 7m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3a8cf71f69d5471cafd2e6eb3a100993 : SUCCESS in 8m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f7e3624d63e649e886eba150de0e0f92 : SUCCESS in 45m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/37e4c1e3873f4de3928fd6eb4f698e9f : FAILURE in 7m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/48570da5d1c24d7cacdd6fd46e98d3bf : SUCCESS in 1h 30m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a8d87cb2cf114e678b4fa642726b96c3 : FAILURE in 1h 22m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2777c647574c4c05a89f5cf71a8e05a8 : SUCCESS in 1h 34m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d2def8c77d864afd9e8045039ba5d937 : SUCCESS in 1h 16m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/30ae2f204a42493db731c4924fa93ba3 : SUCCESS in 1h 18m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5c61c677fd5d43a9840dbb3ab01075a2 : SUCCESS in 1h 24m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/839337d439d8460687113613fc43360b : SUCCESS in 2h 24m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20697be8cecc41c98adf7520640d6a86 : FAILURE in 34m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5eb0d3054b32e1cf4bc3e631ddc8349c68b90b36","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-01-18 14:34:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/05/825105/1/check/iscsi-driver-alletra-6k/ea9f3cd/ea9f3cd.html : FAILURE in 50m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8d2beb209838c421382d507a2e969c8db80b306a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-18 15:31:04.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/825105/1/check/tempest-dsvm-full-focal-py3/10ef04e/ : SUCCESS in 2h 39m 19s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/825105/1/check/devstack-plugin-ceph-dsvm-focal-py3/123ae8f/ : FAILURE in 1h 21m 12s","accounts_in_message":[],"_revision_number":1},{"id":"02914b9d5ea3f732e368dfe2a2491f726fc58bae","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-01-18 17:00:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/825105/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/e9b47b9/ : FAILURE in 2h 51m 05s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/825105/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/e5d320f/ : TIMED_OUT in 3h 02m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6f6ad9cde7ba35cfe46a61aa14b2dabb614d3e87","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-01-18 20:56:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/05/825105/1/ : SUCCESS in 1h 46m 56s","accounts_in_message":[],"_revision_number":1},{"id":"2e60bd2141749c59ee5dd05d919d87cf6deebb04","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-18 21:57:53.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-825105-1/ns5_nfs : SUCCESS in 5h 28m 44s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-05-825105-1/ns5_iscsi : SUCCESS in 5h 6m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"51a9a4b8457e356e3cc55cff0e581dacefc86e10","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-01-18 22:14:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/05/825105/1/check/dsvm-tempest-cisco-zonemanager-job/c8723e2/ : SUCCESS in 1h 14m 48s","accounts_in_message":[],"_revision_number":1},{"id":"fb20a35eb2dfbd9ae401592ae8aa9e315f21f319","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-01-22 10:17:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/825105/1/20220122130830/NEC-FC : SUCCESS in 2h 45m 12s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/825105/1/20220122130830/NEC-ISCSI : SUCCESS in 6h 08m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ac583c603c3ec3aba4b79bfecf4fcb8c1904a5f4","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-01-31 16:30:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"322d5c237fa78faaba257b6b1875b0d80748bb12","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-01-31 17:45:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/ed130cd/index.html : SUCCESS in 1h 12m 05s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/3907890/index.html : SUCCESS in 1h 13m 41s","accounts_in_message":[],"_revision_number":1},{"id":"6afd7c9cc5413c237520e94ae356fd86ba883caf","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-31 18:24:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825105/1/iscsi/ : SUCCESS in 1h 00m 19s\n- cinder-smb http://cloudbase-ci.com/cinder/825105/1/smb/ : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5f28e33670ba1ca59802621e4f94fa6a969808cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 19:55:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/531437a29b1d485b91ae90a973e07fc1 : SUCCESS in 12m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e17d49bfe5945a2b629361b0f83c31a : SUCCESS in 13m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a022a139fb634940b0699bd0de703a13 : SUCCESS in 12m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48547c48b5604b759898fae528a987b4 : SUCCESS in 15m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/10906e645a3c4e17bb066d452c2b3a17 : FAILURE in 12m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f3a2ccfb0fce459f8b5b74b3c7f0f54c : SUCCESS in 59m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99bae8f1b0a14d23947b5a929c329296 : SUCCESS in 47m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/656cf707df974a88ae37a930eb9a61b0 : SUCCESS in 17m 47s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/435e545140f644fbb41999c8111086b1 : SUCCESS in 10m 30s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/15b99f0b57af4fc3b5e7afab38a79819 : SUCCESS in 12m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a033caefbb1a476e9880337ca1956f6b : SUCCESS in 10m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8cc76e85d9684484b4e072ffc1291aac : SUCCESS in 46m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c09a0f4117a24c65a40b6e2b74e830a9 : FAILURE in 10m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/698090de19b64cd2a78f0b7f6c11ef7c : POST_FAILURE in 1h 15m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e011bb98d0ef4e8c9642c47d94a8cf72 : FAILURE in 1h 51m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/71dccee385ba4d10a539d9b22164f8b6 : SUCCESS in 1h 20m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5c4ad8887cb940deb9d184ac4f5cbf90 : FAILURE in 15m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c1e2a41994d14d9e92e84d8e6e8a829c : SUCCESS in 1h 32m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4aeebd9491a148649814872721254b9f : SUCCESS in 1h 36m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ee34c035aaf84c49acae5f1b75bf5bc4 : SUCCESS in 2h 00m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5f5deafd02d4b57a151326178ac54ec : SUCCESS in 1h 25m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b8f7c4584b02dd3d43cd240c0f1cb5e091daec33","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-31 21:08:06.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/825105/1/check/tempest-dsvm-full-focal-py3/1bed20f/ : SUCCESS in 1h 41m 34s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/825105/1/check/devstack-plugin-ceph-dsvm-focal-py3/0bb417f/ : FAILURE in 1h 24m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f0c5e51cf4f4e3b22b88ddcc2e37c55b7d444fd3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-01-31 22:43:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/05/825105/1/check/dsvm-tempest-cisco-zonemanager-job/1dd663e/ : SUCCESS in 1h 14m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b29e05c85ba8349bccba9bdf4012dad7eb253988","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-06 04:56:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45ca98b6dc7afd57c573d37a2fd19572395cd9a1","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-02-09 14:53:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d54c40c4953c99bd15a9c7200ae7660a33cb403","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-09 16:12:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825105/1/iscsi/ : SUCCESS in 1h 06m 19s\n- cinder-smb http://cloudbase-ci.com/cinder/825105/1/smb/ : SUCCESS in 1h 17m 21s","accounts_in_message":[],"_revision_number":1},{"id":"295ca45ae64b4fd780610d9ac44c667c1e18d207","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-02-09 16:59:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/30a19b6/index.html : SUCCESS in 1h 12m 50s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/2ca8b18/index.html : SUCCESS in 1h 04m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8a8fb31abd4b671f9da4a4cee2db89a8f7607e6a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-09 17:19:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/05/825105/1/check/dsvm-tempest-cisco-zonemanager-job/d567485/ : SUCCESS in 1h 15m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a72054fd1e2ceff4310ec4f0221c05c16374d761","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-09 17:40:38.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/825105/1/check/tempest-dsvm-full-focal-py3/7aa4457/ : SUCCESS in 2h 23m 21s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/05/825105/1/check/devstack-plugin-ceph-dsvm-focal-py3/3484488/ : FAILURE in 1h 26m 32s","accounts_in_message":[],"_revision_number":1},{"id":"6cae0f5957f184a5218e5c1d31c91857e2d75756","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 17:44:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e84dd692c2b4432091029bcf31461c07 : SUCCESS in 10m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f4983ddab23407da98b2ca2d5ea0da5 : SUCCESS in 12m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa67f88e69fd4c1f93e13387db218a3e : SUCCESS in 13m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d8eb54e31ef4f969124963d7dec4a8b : SUCCESS in 17m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/2f6b20792133411fa92cacaa82db0985 : SUCCESS in 1h 01m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/57ef240b24ad41879e36371f48e4dba8 : SUCCESS in 1h 04m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdc9a6d5a6d94040aff78f5c31ce9cac : SUCCESS in 54m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4751d22cfc1249aebb77924bc0b5382f : SUCCESS in 15m 00s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7d3019e99df54d38bdedd4b15081342a : SUCCESS in 11m 04s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cbaf7d9d0a014240a17d31ce06d3cb41 : SUCCESS in 9m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/058849df59344a0c9c69e88ae7092bb3 : SUCCESS in 9m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7c0b9688e38841bcaf582b9befbbbd4b : SUCCESS in 44m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a3dda5637899462ea464b2a05d384c7f : FAILURE in 9m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1b99fc5f87754f4bbe56eef572e08b40 : SUCCESS in 1h 40m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/00fdc6d08c0a4b17bcba13bc2797759d : FAILURE in 2h 01m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/671dddff9f7d4b7683cf48eaa4eb52bb : SUCCESS in 1h 36m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b77fbbcee1b348748a2e3ac6cccfd8fa : SUCCESS in 1h 12m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d5c84f961a2644f88459540f22e3790a : SUCCESS in 1h 20m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/75018707519046f79b37764b97088975 : SUCCESS in 1h 24m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5847041077d04708a3eb56666a66a770 : SUCCESS in 2h 17m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/caff4568846543ed8e9c1a22c598ac8c : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":1},{"id":"090f66f1300e13c4a213bf20e4afc62cfc5bdad4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-10 13:30:37.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f305e6d3010849d8c3e26be3f285eac9cf658fae","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-02-10 14:46:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"abcc55f1a6641e9c29ab2f20b9af5b3b8a422321","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 14:47:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39c2376ef1478522dc45141bff860db62aeaf384","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 17:00:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14936593c56c4e6db536922684cdd410 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/027151ee778d42589a8e19798d30c714 : SUCCESS in 10m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7021b8fb5af545679531e9ba29091afa : SUCCESS in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16edb3712d6547428a4cd1aec3dfaa42 : SUCCESS in 16m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/d7ae0c8145ab4477a9d01c848b490f06 : SUCCESS in 59m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/73919423e7b34b2d90269ae4d78f8784 : SUCCESS in 1h 23m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b238cb88e75d4e7fa7a2f2a18e7cdb15 : SUCCESS in 58m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cca8f44178ec41fd97e8403488188e64 : SUCCESS in 1h 28m 19s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/458cabd30a214a37a1513dc65881390c : SUCCESS in 1h 37m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/068dba1b4c194cb185c6fb25e39a42e8 : SUCCESS in 56m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a104d9ae714296255f855778fed8934d07abaf7a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 17:00:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"949a292ebfee6affb7aa6e75af37c7aac3c99cf8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 17:12:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f6faba8add9465087dc5240cd8f51e4 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc6585577ee06825dc72a82f815069cba6372fa6","revisions":{"bc6585577ee06825dc72a82f815069cba6372fa6":{"kind":"REWORK","_number":1,"created":"2022-01-18 11:54:40.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/05/825105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/825105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/825105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/825105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/825105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/825105/1"}}},"commit":{"parents":[{"commit":"3339bfb83963e0d60733696ed62462e8590b961d","subject":"Fix volume OVO create method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3339bfb83963e0d60733696ed62462e8590b961d"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-03 14:55:33.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-01-18 11:11:31.000000000","tz":60},"subject":"Expose volume_attachments in Volume OVO","message":"Expose volume_attachments in Volume OVO\n\nIn change-id Iabf9c3fab56ffef50695ce45745f193273822b39 we left the\n`volume_attachment` out of the expected attributes of the Volume OVO\n(even when te DB layer is providing that information) because it was\nbreaking some unit tests.\n\nThis means that in some cases we are unnecessarily loading the\nattachments again (manually or via lazy loading) once we have the volume\nOVO because that field is not set.\n\nIn this patch we populate the `volume_attachment` when we load the\nVolume OVO from the DB.\n\nChange-Id: I6576832b2c2722ab749cfe70bbc2058ead816c36\n(cherry picked from commit e07bf3787a7c1a8a55b8f69290cf6166719f6a5e)\nConflicts:\n\tcinder/objects/volume.py\n\tcinder/tests/unit/objects/test_volume.py\n(cherry picked from commit e85c22cef7bd502db44f837e626b7262f4dc8241)\n(cherry picked from commit a5b67fd959fb837120745e1a2757f18060dd6ecc)\nConflicts:\n\tcinder/tests/unit/volume/flows/test_create_volume_flow.py\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc6585577ee06825dc72a82f815069cba6372fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc6585577ee06825dc72a82f815069cba6372fa6"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
