)]}'
{"id":"openstack%2Fcinder~825492","triplet_id":"openstack%2Fcinder~master~I586b79519573add80a1c34d9ba2656073852b4f7","project":"openstack/cinder","branch":"master","topic":"bp/yadro-tatlin-unified-driver","attention_set":{"34497":{"account":{"_account_id":34497,"name":"kid1972","display_name":"Sergey Karpenkov","email":"s.karpenkov@yadro.com","username":"yadro"},"last_update":"2022-07-28 14:45:59.000000000","reason":"\u003cGERRIT_ACCOUNT_13671\u003e replied on the change","reason_account":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"}}},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2022-07-28 10:26:46.000000000","reason":"Change was submitted"},"34465":{"account":{"_account_id":34465,"name":"Peha","email":"a.pavlov@yadro.com","username":"apavlov1992"},"last_update":"2022-05-23 11:02:19.000000000","reason":"Raghavendra Tilay replied on the change"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-07-27 11:00:57.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2022-06-07 06:07:02.000000000","reason":"Raghavendra Tilay replied on the change"},"33746":{"account":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"last_update":"2022-06-06 13:40:25.000000000","reason":"removed on reply"},"31779":{"account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"last_update":"2022-07-12 14:45:32.000000000","reason":"\u003cGERRIT_ACCOUNT_31779\u003e replied on the change","reason_account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"}},"13671":{"account":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"last_update":"2022-07-28 10:26:46.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2022-07-28 10:26:46.000000000","reason":"Change was submitted"},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2022-07-28 10:26:46.000000000","reason":"Change was submitted"},"33437":{"account":{"_account_id":33437,"name":"YADRO","email":"partners@yadro.com"},"last_update":"2022-05-11 11:24:54.000000000","reason":"Vladislav Belogrudov replied on the change"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-07-28 08:48:14.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I586b79519573add80a1c34d9ba2656073852b4f7","subject":"Initial commit for Yadro Tatlin.UNIFIED driver","status":"MERGED","created":"2022-01-20 12:30:39.000000000","updated":"2022-07-28 17:21:05.000000000","submitted":"2022-07-28 10:26:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":152,"unresolved_comment_count":15,"has_review_started":true,"submission_id":"825492-bp/yadro-tatlin-unified-driver","meta_rev_id":"cb7671dbf17e45c383cba3cc6221f61a6379e4a8","_number":825492,"virtual_id_number":825492,"owner":{"_account_id":34497,"name":"kid1972","display_name":"Sergey Karpenkov","email":"s.karpenkov@yadro.com","username":"yadro"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-28 10:26:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":2,"date":"2022-07-27 11:00:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":2,"date":"2022-07-28 08:48:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":1,"date":"2022-07-28 08:48:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":1,"date":"2022-07-27 11:00:57.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34465,"name":"Peha","email":"a.pavlov@yadro.com","username":"apavlov1992"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"}],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":34465,"name":"Peha","email":"a.pavlov@yadro.com","username":"apavlov1992"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-20 12:30:50.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-20 12:55:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-20 13:29:44.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-20 13:35:51.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-20 14:08:40.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-01-20 15:01:51.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-01-20 16:56:05.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-01-20 17:52:32.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-20 17:55:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-20 18:05:06.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-01-20 18:09:23.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-01-20 18:11:03.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-20 18:39:28.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-01-20 20:14:03.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2022-01-20 21:54:51.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-21 15:21:12.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-25 07:57:28.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-11 11:24:54.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"state":"REVIEWER"},{"updated":"2022-05-12 10:30:13.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2022-05-19 15:27:11.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-05-19 16:20:57.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-05-19 16:30:50.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-19 17:25:22.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-05-19 17:36:11.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-05-19 22:34:31.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-20 01:04:48.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-05-20 01:12:31.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-05-20 06:55:04.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-05-20 07:00:22.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":34465,"name":"Peha","email":"a.pavlov@yadro.com","username":"apavlov1992"},"state":"REVIEWER"},{"updated":"2022-05-20 08:23:05.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-05-20 12:37:55.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-05-22 09:03:37.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-05-23 18:10:19.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-05-24 11:52:04.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2022-05-26 20:00:05.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-05-31 04:50:41.000000000","updated_by":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"reviewer":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"state":"CC"},{"updated":"2022-06-08 14:49:30.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2022-06-17 12:20:16.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2022-06-17 15:02:10.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"CC"},{"updated":"2022-06-19 07:59:13.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-27 03:00:19.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2022-07-11 08:28:16.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-11 18:32:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2022-07-12 21:20:31.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-07-13 12:04:22.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"4e2ff4afb5d1f2cee0cf76ab4399f967ea0d2e78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34497,"name":"kid1972","display_name":"Sergey Karpenkov","email":"s.karpenkov@yadro.com","username":"yadro"},"date":"2022-01-20 12:30:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bbb683d82a22def194c72997f2d19cc4d9bacbc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-20 12:30:50.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/14990/","accounts_in_message":[],"_revision_number":1},{"id":"c94ef4afbee0a976e1dcf7c69f04f3164577f17e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-20 12:32:40.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/14990/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f6a1a5e21e4173159dac797553e0037a06321bf3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 12:55:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8cdc26bf0ec04eb8b7fd6a994ad6dc3c : SUCCESS in 14m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/868215c02275444b9712149861bebf70 : SUCCESS in 16m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8d80c30ca37699b1ca501e3086a28c1420235e09","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-20 13:25:24.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/825492/825492/1/check-not-vote/ext-cinder-zuul/66f6b0c : SUCCESS in 33m 22s","accounts_in_message":[],"_revision_number":1},{"id":"6247586d36922e67aefa7f088f6a77c1a4cd9bd8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-20 13:29:44.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/1/check/tempest-dsvm-full-focal-py3/b7c2557/ : FAILURE in 49m 51s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/1/check/devstack-plugin-ceph-dsvm-focal-py3/bc539e6/ : FAILURE in 53m 25s","accounts_in_message":[],"_revision_number":1},{"id":"37ffaf2a759852c459d8587ad77f66d707e864ba","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-01-20 13:35:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/92/825492/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d25d5c6/d25d5c6.html : FAILURE in 1h 02m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c0f9f4bd9d03eb08e8a72517991cbf1250309c82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34497,"name":"kid1972","display_name":"Sergey Karpenkov","email":"s.karpenkov@yadro.com","username":"yadro"},"date":"2022-01-20 13:40:19.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6e7a1e1d21afc3231efbea108f590b5b7fc0ef7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33437,"name":"YADRO","email":"partners@yadro.com"},"date":"2022-01-20 13:53:54.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"efdc9fc62e8ba25ed43c20181e4ab9771fb86c2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33437,"name":"YADRO","email":"partners@yadro.com"},"date":"2022-01-20 13:54:59.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"44778f6767bda4ba0c26a821076953d8430c08d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33437,"name":"YADRO","email":"partners@yadro.com"},"date":"2022-01-20 13:55:28.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"717c43b3643fd06b1c5285037cbf2519ec77f4d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33437,"name":"YADRO","email":"partners@yadro.com"},"date":"2022-01-20 13:57:37.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"47d9af99ecfdfd3c7f1f469268509c05f0eacd8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33437,"name":"YADRO","email":"partners@yadro.com"},"date":"2022-01-20 14:01:16.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"941dc410bb70c430f8ff81033b1da6ba3cf4f23d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33437,"name":"YADRO","email":"partners@yadro.com"},"date":"2022-01-20 14:02:34.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"928e09843fb5aeb17bb23b7017f0d3a11194c033","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 14:08:40.000000000","message":"Patch Set 4:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-test/23/ : FAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/23 \n\nnull","accounts_in_message":[],"_revision_number":4},{"id":"284c39b2aa8f7f4acdc930b6621effb923a73428","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 14:08:41.000000000","message":"Patch Set 5:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-test/24/ : FAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/24 \n\nnull","accounts_in_message":[],"_revision_number":5},{"id":"a634273b46e73973f81dde44156f9ccba3d9ec5a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 14:08:42.000000000","message":"Patch Set 3:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-test/22/ : FAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/22 \n\nnull","accounts_in_message":[],"_revision_number":3},{"id":"7839f32c1e6464bbcccf6c22b6a51082bba896f1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 14:08:43.000000000","message":"Patch Set 6:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-test/25/ : FAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/25 \n\nnull","accounts_in_message":[],"_revision_number":6},{"id":"6ae871f85ebfce24dd9f2007fdefdff4881f006e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 14:08:46.000000000","message":"Patch Set 8:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-test/27/ : FAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/27 \n\nnull","accounts_in_message":[],"_revision_number":8},{"id":"1d351ba37c1302bcd9d60188590067de8b78f95c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 14:08:50.000000000","message":"Patch Set 7:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-test/26/ : FAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/26 \n\nnull","accounts_in_message":[],"_revision_number":7},{"id":"510e35331f49f7ef0e10d8f9647fa5be24656ed4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 14:20:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c197a43a6ec84f7f9cea185089c0326a : SUCCESS in 16m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/966cae537b134fccb10ea4b1d7c7e60e : SUCCESS in 16m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1e445dfa42ac70c7ff49563909ff68ad0d59ceda","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-20 15:01:51.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/8/iscsi/ : SUCCESS in 58m 46s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/8/smb/ : SUCCESS in 50m 47s","accounts_in_message":[],"_revision_number":8},{"id":"a616676d6236879038530a14044c49933d09d66f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-20 15:03:16.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/8/check/tempest-dsvm-full-focal-py3/76177c0/ : FAILURE in 50m 27s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/8/check/devstack-plugin-ceph-dsvm-focal-py3/3f8adb4/ : FAILURE in 53m 05s","accounts_in_message":[],"_revision_number":8},{"id":"05498835acbb800250cb8d557849bf75eb6d89de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34497,"name":"kid1972","display_name":"Sergey Karpenkov","email":"s.karpenkov@yadro.com","username":"yadro"},"date":"2022-01-20 15:07:25.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7119186f42bf3750d9b71cc1b3251747e971c933","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34497,"name":"kid1972","display_name":"Sergey Karpenkov","email":"s.karpenkov@yadro.com","username":"yadro"},"date":"2022-01-20 15:17:07.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"db8ecc706d1cdaa00a90f4575271d2d22af3adbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 15:53:47.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4cce9913c6724a4eb16ea98b41ea8ab1 : SUCCESS in 20m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0e7f465115f47f1b5fa20f1013896d9 : SUCCESS in 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e5f5e44273085171139777cb9742d36875cb648a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 16:21:42.000000000","message":"Patch Set 10:\n\nFAILED \n\nBuild log link http://tempest-openstack.spb.yadro.com/gerrit-test/31  : FAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/31 \n\nnull","accounts_in_message":[],"_revision_number":10},{"id":"5eff3776b4389fca1b3ead8640f74e68b664fb79","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 16:22:16.000000000","message":"Patch Set 10:\n\nFAILED \n\nBuild log link http://tempest-openstack.spb.yadro.com/gerrit-test/32  : FAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/32 \n\nnull","accounts_in_message":[],"_revision_number":10},{"id":"5185c33cedb343cd40f883c5675525340ded66d0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 16:22:22.000000000","message":"Patch Set 10:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":10},{"id":"bb43a9e1fcc2d3d5e577fe194b0f7bf427b1db58","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 16:22:26.000000000","message":"Patch Set 10:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":10},{"id":"54561a2567c56d82348109080c1f72b5630076a0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 16:22:30.000000000","message":"Patch Set 10:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":10},{"id":"46d7794af44a46dae64396b12fdbca34cb10e7b6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 16:22:34.000000000","message":"Patch Set 10:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":10},{"id":"9554e132b2529619192f26fa3f2af28de4e7496f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 16:22:45.000000000","message":"Patch Set 10:\n\nFAILED \n\nBuild log link http://tempest-openstack.spb.yadro.com/gerrit-test/33  : FAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/33 \n\nnull","accounts_in_message":[],"_revision_number":10},{"id":"ab176346ce42a1e480d29fae0c6d9ae6c592ef9d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 16:23:51.000000000","message":"Patch Set 10:\n\nFAILED \n\nBuild log link http://tempest-openstack.spb.yadro.com/gerrit-test/34  : ABORTED\n\nnull","accounts_in_message":[],"_revision_number":10},{"id":"592bbbdd60c9bd9c31d8481ae701cad3b72a4ab1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 16:23:55.000000000","message":"Patch Set 10:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":10},{"id":"5f3abd4ac8fe4ecdebf75e2eac82403f29182b32","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 16:24:09.000000000","message":"Patch Set 10:\n\nFAILED \n\nBuild log link http://tempest-openstack.spb.yadro.com/gerrit-test/35  : ABORTED\n\nnull","accounts_in_message":[],"_revision_number":10},{"id":"736fc141b3e03316c0c4028415a5283bd4fcf4c2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 16:24:37.000000000","message":"Patch Set 10:\n\nFAILED \n\nBuild log link http://tempest-openstack.spb.yadro.com/gerrit-test/36  : ABORTED\n\nnull","accounts_in_message":[],"_revision_number":10},{"id":"007f683d36209aacd32096118ac309667dcd947e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-01-20 16:24:43.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/92/825492/10/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/aadc207/aadc207.html : FAILURE in 1h 02m 52s","accounts_in_message":[],"_revision_number":10},{"id":"5ab5599d01c66601f7df2b35c8138de0b37de723","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 16:24:52.000000000","message":"Patch Set 10:\n\nFAILED \n\nBuild log link http://tempest-openstack.spb.yadro.com/gerrit-test/37  : ABORTED\n\nnull","accounts_in_message":[],"_revision_number":10},{"id":"e46cf34bb4937427a5d2a5dcaabf79575a7a3d62","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 16:25:30.000000000","message":"Patch Set 10:\n\nFAILED \n\nBuild log link http://tempest-openstack.spb.yadro.com/gerrit-test/39  : ABORTED\n\nnull","accounts_in_message":[],"_revision_number":10},{"id":"18705875e6d7df94bdd656f60b4d0657f5f44ff3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 16:25:35.000000000","message":"Patch Set 10:\n\nFAILED \n\nBuild log link http://tempest-openstack.spb.yadro.com/gerrit-test/38  : ABORTED\n\nnull","accounts_in_message":[],"_revision_number":10},{"id":"5c44d2fe429e518d49bf0983bd34c0c0310d4f82","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-01-20 16:43:34.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/10/iscsi/ : FAILURE in 1h 25m 57s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/10/smb/ : SUCCESS in 1h 06m 25s","accounts_in_message":[],"_revision_number":10},{"id":"be9f6d0b0053068d37ba57a74f0e34b4397409cd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 16:50:16.000000000","message":"Patch Set 10:\n\nSUCCESSFUL \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-test/44/ : SUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-test/44 \n\nnull","accounts_in_message":[],"_revision_number":10},{"id":"189d28d9cbf926119e6936658a44c9ce55a95862","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-01-20 16:56:05.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/92/825492/10/check/iscsi-driver-alletra-6k/abf0681/abf0681.html : SUCCESS in 1h 37m 05s","accounts_in_message":[],"_revision_number":10},{"id":"b551788caa877c71f228759e6d46f3de24b6e712","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-20 17:05:01.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/10/check/tempest-dsvm-full-focal-py3/2367194/ : FAILURE in 1h 08m 07s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/10/check/devstack-plugin-ceph-dsvm-focal-py3/c838818/ : FAILURE in 1h 11m 13s","accounts_in_message":[],"_revision_number":10},{"id":"7c7f033a832dd284cfa26d3dc70c32d7dac6e92a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-01-20 17:30:13.000000000","message":"Patch Set 10:\n\nSUCCESSFUL \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-test/45/ : SUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-test/45 \n\nnull","accounts_in_message":[],"_revision_number":10},{"id":"5d8d9e472d4fef0fbd1505721d732a084ee71149","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-20 17:52:32.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-1/ns5_nfs : SUCCESS in 4h 19m 5s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-1/ns5_iscsi : SUCCESS in 4h 27m 20s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"4c3530353b5486eb5c63b5374244e6dccccac8d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 17:55:07.000000000","message":"Patch Set 10: Verified-1\n\n(356 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dc9c4cc979f41809db3785cc4b01914 : FAILURE in 14m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e11f287a10ec45b7b7566ea0ae375faf : SUCCESS in 11m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57df8daf73d94074b6f5db1842d2e8c2 : SUCCESS in 11m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42723aa9228d4323980dbae05fdf3c67 : SUCCESS in 13m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/763be06f545c43a8a41f9fe3afbfeaf2 : SUCCESS in 56m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cff7e203a72d490a9948963ee6db02ec : SUCCESS in 1h 26m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c4de5309b9c4c52bfc25e7e67965f1c : SUCCESS in 48m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2c9b5b269fd9484397ca6de83653cb90 : SUCCESS in 16m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d282e86480584dd78257960234bd4f90 : FAILURE in 9m 03s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/468da085a9ce419d9c248d29b7f4fe45 : SUCCESS in 9m 28s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/11b15f6dca134ddabca36b2cd921ea3f : SUCCESS in 9m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/edab7b97777d4fd282c76dd8e3f6608b : SUCCESS in 9m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c2695328d60240a1a42d4d91f9a370f6 : SUCCESS in 49m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a2d39269c8054f63ab07e93155285bbc : FAILURE in 10m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4af9a27e2c92496893b4a4c2295bca07 : SUCCESS in 1h 15m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/97e64b912d1b4fa086f68b37e93b4d91 : SUCCESS in 1h 29m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d26bf2ecaed8423a8bbf3e3b9b30cd7d : SUCCESS in 1h 38m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c677c469f8274ced91e199564ba76e9d : SUCCESS in 1h 08m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/76edb7c315f34842bb7674a07097d4cc : SUCCESS in 1h 30m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1fe190110dbc4420863d2534c8390b99 : SUCCESS in 41m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/914406cd124c464293bcc11efc8e41a2 : SUCCESS in 1h 36m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/11c48fcb20004cf1b156648dc70a1627 : SUCCESS in 2h 30m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24dabb58cd9a467494b6358611c5d052 : SUCCESS in 45m 00s","accounts_in_message":[],"_revision_number":10},{"id":"286f54599817c8ce28fbf5657c9d0e93864127c4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-01-20 18:05:06.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/92/825492/10/check/dsvm-tempest-cisco-zonemanager-job/527f929/ : SUCCESS in 1h 07m 25s","accounts_in_message":[],"_revision_number":10},{"id":"ec98b643f273f57cb49d3c58f595a8bb621e83fa","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-01-20 18:09:23.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/10/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/0d79005/ : FAILURE in 2h 50m 32s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/825492/10/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3bd09e4/ : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":10},{"id":"b423ad96f209c2054ee5dc1a8bab8afbaf48cecb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-01-20 18:11:03.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b65f2329c731439cbe0b424791cb5c6c : FAILURE in 1h 25m 17s","accounts_in_message":[],"_revision_number":10},{"id":"cb7f58fcb79754454b944f8804fd0b8c59116e76","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-20 18:39:28.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/825492/10/check/DellEMC_Unity_FC/05c161d/DellEMC_Unity_FC/437 : SUCCESS in 1h 11m 29s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/10/check/DellEMC_Unity_iSCSI/cd98154/DellEMC_Unity_iSCSI/429 : SUCCESS in 1h 27m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"be22bb6234476221a644d6ad7064350ee6ff295b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-20 19:32:20.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-2/ns5_nfs : SUCCESS in 4h 30m 18s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-2/ns5_iscsi : SUCCESS in 4h 37m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"34449c88c5e1bdca7de8b11b5fecdc2d24179196","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2022-01-20 20:14:03.000000000","message":"Patch Set 10:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU ACS5000 CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_ACS5000_CI\n\n- dsvm-tempest-TOYOU-ACS5000-driver http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/92/825492/10/check/dsvm-tempest-TOYOU-ACS5000-driver/2605c72 : SUCCESS in 1h 54m 21s\n- fc-test http://ec2-13-112-122-9.ap-northeast-1.compute.amazonaws.com:9090/92/825492/10/check/fc-test/f80669f : SUCCESS in 2h 02m 21s","accounts_in_message":[],"_revision_number":10},{"id":"d8badd8335e80a52eafaec52780177637a04adc3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-20 20:20:27.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-3/ns5_nfs : SUCCESS in 4h 33m 36s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-3/ns5_iscsi : SUCCESS in 4h 40m 7s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"b766b4ad9909d7367b43ecbc63c13a222aa0e28b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-20 21:05:53.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-4/ns5_nfs : SUCCESS in 4h 36m 8s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-4/ns5_iscsi : SUCCESS in 4h 40m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"2c43ba758e42f909fa662964baaa2c234f10a865","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-01-20 21:54:51.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/92/825492/10/check/DellEMC_XtremIO_FC/4a8d369/DellEMC_XtremIO_FC/522 : SUCCESS in 57m 41s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/10/check/DellEMC_XtremIO_iSCSI/76db3c7/DellEMC_XtremIO_iSCSI/532 : SUCCESS in 1h 45m 05s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":10},{"id":"6814392ad3bc278ad884fc9d1f83eb62542f5cb5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-20 22:36:24.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-5/ns5_nfs : SUCCESS in 4h 41m 4s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-5/ns5_iscsi : SUCCESS in 4h 47m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"71b100a7b50f798a7364bbba47afbfa371c8e35a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-20 22:43:16.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-6/ns5_nfs : SUCCESS in 4h 38m 0s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-6/ns5_iscsi : FAILURE in 4h 47m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"b37e4159cf7e906717b91530a7e2ff9dce7be242","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-21 00:06:05.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/825492/825492/10/check-not-vote/ext-cinder-zuul/1d3669b : SUCCESS in 31m 15s","accounts_in_message":[],"_revision_number":10},{"id":"29cd1e747d149cca5c5f7d169187ddcfd10f6492","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-21 00:18:54.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-7/ns5_nfs : SUCCESS in 4h 39m 58s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-7/ns5_iscsi : SUCCESS in 4h 45m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"fd90731663ce5259f2f4b0239f5fc9b9058d4707","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-21 01:15:05.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-8/ns5_nfs : SUCCESS in 4h 40m 51s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-8/ns5_iscsi : SUCCESS in 4h 47m 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"2a0c1ebe204f59fb021a1549a2f56f95e6966091","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-21 01:35:21.000000000","message":"Patch Set 10:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":10},{"id":"0c34bfab5d3a138e0e26e6c8954cdbe97aa42c20","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-21 01:35:39.000000000","message":"Patch Set 9:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":9},{"id":"e1e3a159e7d95ddd7b3b8828ab67c105387354e4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-21 03:26:48.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-9/ns5_nfs : SUCCESS in 4h 52m 8s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-9/ns5_iscsi : SUCCESS in 4h 46m 22s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"bb1b0a9e95d1417816d2ce0e4fd0f542092546d6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-21 05:11:00.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-10/ns5_nfs : SUCCESS in 4h 52m 27s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-825492-10/ns5_iscsi : SUCCESS in 4h 49m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"91df10d221d8234196235f7fe62c04a4cf6a4c92","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-01-21 15:21:12.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/92/825492/10/ : RETRY_LIMIT in 2m 09s","accounts_in_message":[],"_revision_number":10},{"id":"36d70ab33573c9a0b07669f82946fccd1a8b6215","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-01-25 07:57:28.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/10/20220125103512/NEC-ISCSI : SUCCESS in 6h 08m 34s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/10/20220125103512/NEC-FC : SUCCESS in 6h 22m 12s","accounts_in_message":[],"_revision_number":10},{"id":"3a11f96377497ca9b39c19f773dc969727d57326","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-05-12 10:30:13.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2be799cb7bb312ec101a9f4c70c524913e0715ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-05-19 13:56:37.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"2e6ec078abf6f41666a3defc5572255af62b64e0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-19 13:56:45.000000000","message":"Patch Set 11:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16045/","accounts_in_message":[],"_revision_number":11},{"id":"3b4aa415ed3a5db29c4e1e1b71f1d7a1f062d1d9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-19 13:58:29.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":11},{"id":"be2f254f710306339d176ea20f7061ebdc560464","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-19 13:58:48.000000000","message":"Patch Set 11:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16045/ : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"ca18b1480dc59d7b9d2cb3215d4277b7ad41c738","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-19 14:01:15.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":11},{"id":"1f49f7c2529cd923defe8e8adffc733fe668a1b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-05-19 14:03:09.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"eb855b47d8d3a2faf626e04984715f465b80c05c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-19 14:05:00.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"c1807d2a4d22918c6b0eea89262ae86aca6eaeae","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-19 14:08:23.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"5f025f295a587ecf6d589d5d7d0ae7ad4fd9149d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 14:55:34.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77d00e72e0254f82882b33a75d24dbf2 : SUCCESS in 24m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b7bd5611d6945a3983e8ab210d80b1e : SUCCESS in 23m 02s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f1c341a0315a0adc0ad461273ae6b0719254f0e2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-19 15:26:02.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/12/iscsi/ : SUCCESS in 1h 08m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/12/smb/ : FAILURE in 1h 04m 23s","accounts_in_message":[],"_revision_number":12},{"id":"c349561439ec547f6074ebd0f2e58e24fc727f73","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-19 15:27:11.000000000","message":"Patch Set 11:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1652968606061 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"60b6bb7f1d0f6960f88ee7637e783937069d9cb4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-19 15:55:12.000000000","message":"Patch Set 12:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1652968996085 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"cceb9e5e69a88f57a197f10340c0e252c5fa8732","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 16:20:49.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ef7266a29384a9fab07ef3ccd2bb36f : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3667e4c0054c420a9b5fd067de73db03 : SUCCESS in 11m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61a8aedf013c42a899673ce0a27e9ff8 : SUCCESS in 11m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fc5059a0662415ca053274afe86eeda : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4667259031a74fc1af32ada2d83314d8 : SUCCESS in 13m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/81386a24129f4ad6beb9442747d4cd91 : SUCCESS in 1h 00m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea6553c315964126aa11e45d6cfc080d : SUCCESS in 1h 31m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72dd32a7fce9415c8e4efa991d9c9b9b : SUCCESS in 1h 02m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/513932de54aa45cba54290855d1558d3 : SUCCESS in 11m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/845d2258cd2b4192ac51b761041ff11f : SUCCESS in 17m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b5030110a11b4096a14b7fd2af11eb99 : SUCCESS in 6m 19s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4c5d07be651948208fa01b5ab7460c46 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/752ba485d9be493c965f85a137542d58 : SUCCESS in 7m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ab823b854dac413e836e587ad609972e : SUCCESS in 7m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fd039f9717074e52b26d445cc59ec3c1 : SUCCESS in 41m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a1cd0d710cee43eea014f2c006e2572a : FAILURE in 7m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/98dda8073d4149809189a5d2b43add04 : SUCCESS in 1h 13m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f9e262001d9e475e93e92ff77805db57 : FAILURE in 1h 41m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/33864b147a374e9cb2d0154b00096423 : SUCCESS in 1h 41m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f5d157f423aa4467b6b36033a011722e : SUCCESS in 1h 07m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ed69094aebca4420be81d777c3599d89 : SUCCESS in 1h 32m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4fd2c44d663c4b5393bd84f685107709 : SUCCESS in 45m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3b9027be26984f6aa2f7677b03af0aef : SUCCESS in 1h 29m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/76a7f7909dfd4cefb107f38f5159bab8 : SUCCESS in 2h 13m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/757a71a5ab064b4984ff9cedaaaf0521 : SUCCESS in 56m 06s","accounts_in_message":[],"_revision_number":12},{"id":"bce706a7d89bfcc5e2bc0f96a69fc47a953b528f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-19 16:20:57.000000000","message":"Patch Set 12:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":12},{"id":"978a531ad03f925af0c79a503abb8c2ddc3fdcf4","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-19 16:30:50.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/12/check/DellEMC_VNX_iSCSI/beae588/DellEMC_VNX_iSCSI/426 : FAILURE in 1s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/92/825492/12/check/DellEMC_VNX_FC/4be0729/DellEMC_VNX_FC/445 : FAILURE in 9m 43s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":12},{"id":"c16e0978f7b33d412da2bb3585680926d2be2813","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-19 16:38:11.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/825492 : SUCCESS in 1h 20m 15s","accounts_in_message":[],"_revision_number":12},{"id":"f913ac10911527f5eee6dd5837afe36cc3774998","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-05-19 17:25:22.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/12/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/32facfc/ : SUCCESS in 1h 01m 58s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":12},{"id":"1d61fb8a7330dc645beb07db88a80290ac5096a5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-05-19 17:36:11.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/12/upstream-check/cinder-tempest-plugin-ontap-fcp/759be1a/ : SUCCESS in 1h 12m 58s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/12/upstream-check/cinder-tempest-plugin-ontap-iscsi/98a3556/ : FAILURE in 1h 06m 26s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/12/upstream-check/cinder-tempest-plugin-ontap-nfs/27bcbec/ : FAILURE in 47m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"4fea8b913c42a2f0c6ad732a5d1d0f8dcd46ee49","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-19 18:10:43.000000000","message":"Patch Set 12:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/7ecbd065e112452ca4d3817d1ff52b8f : POST_FAILURE in 2h 43m 14s","accounts_in_message":[],"_revision_number":12},{"id":"9c8b37cb2041d27ede82f2bf85a68caee22a050b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-19 18:14:12.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-11/ns5_nfs : SUCCESS in 3h 33m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-11/ns5_iscsi : SUCCESS in 3h 35m 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"f1b2efef112386f334e07a3f3d4233431fe4ad92","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-19 18:41:47.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-12/ns5_nfs : SUCCESS in 3h 34m 31s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-12/ns5_iscsi : SUCCESS in 3h 40m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"ecf95fb81997c72a080253bc9ac92598a4283f91","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-19 18:47:21.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/12/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/7dab21f/ : TIMED_OUT in 3h 03m 28s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/825492/12/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ebcaaa0/ : FAILURE in 1h 12m 12s","accounts_in_message":[],"_revision_number":12},{"id":"706f490054a6f63145d5b5739918ef7bf044cf0d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-19 22:34:31.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/92/825492/12/check/DellEMC_PowerFlex/bcf4ee6/DellEMC_PowerFlex/770 : SUCCESS in 1h 15m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"74697d3365c08b8d9e70f1745ffb9f22e6742e46","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-20 00:18:18.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/825492/12/check/iscsi-driver-nimble/3db651e/3db651e.html : SUCCESS in 1h 35m 59s","accounts_in_message":[],"_revision_number":12},{"id":"c829884daf4db691bea37ad509ce363190b27d3d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-20 01:00:02.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/825492/12/check/DellEMC_Unity_FC/db9f0be/DellEMC_Unity_FC/799 : SUCCESS in 1h 11m 41s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/12/check/DellEMC_Unity_iSCSI/6bd8d7c/DellEMC_Unity_iSCSI/767 : SUCCESS in 1h 39m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"99df5820029a4769f20c8c0b5225b8c11fb51ea9","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-05-20 01:04:48.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/825492-12/HPMSA-iSCSI/ : SUCCESS in 58:49.91\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/825492-12/HPMSA-FC/ : SUCCESS in 56:22.25","accounts_in_message":[],"_revision_number":12},{"id":"bc24a67ea47e853da54132fbd456fa8bf563ff78","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-05-20 01:12:31.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/92/825492/12/check/DellEMC_SC_FC/171346e/DellEMC_SC_FC/616 : SUCCESS in 50m 47s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/12/check/DellEMC_SC_iSCSI/de55fcb/DellEMC_SC_iSCSI/637 : FAILURE in 1h 35m 09s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":12},{"id":"1efcabe8d676f966df097dde6d175993b3783c3e","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-20 06:55:04.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/825492-12/Lenovo-iSCSI/ : SUCCESS in 1:05:26\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/825492-12/Lenovo-FC/ : SUCCESS in 1:02:33","accounts_in_message":[],"_revision_number":12},{"id":"525aa4e77af1c033e3bf1db890b3830240a42f68","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-05-20 08:23:05.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/825492-12/PVME-iSCSI/ : SUCCESS in 1:03:35\n- PVME-FC https://os-logs.seagate.com/dell-ci/825492-12/PVME-FC/ : SUCCESS in 1:02:13","accounts_in_message":[],"_revision_number":12},{"id":"2ae8dea2e4761c19f31001b67d8960f55e9ab70a","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-05-20 12:37:55.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/825492-12/Seagate-iSCSI/ : SUCCESS in 1:05:51\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/825492-12/Seagate-FC/ : SUCCESS in 1:05:04","accounts_in_message":[],"_revision_number":12},{"id":"bdc8b6dcfa752c08fb650daf65285ce43fcbd6f2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-22 09:03:37.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/12/20220522113917/NEC-V-FC : SUCCESS in 5h 42m 48s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/12/20220522113917/NEC-V-ISCSI : SUCCESS in 6h 24m 17s","accounts_in_message":[],"_revision_number":12},{"id":"119ef676b006169cd8c8a2cf1e78a9fbd5e5c395","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-05-22 19:04:35.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/12/20220522211149/NEC-FC : SUCCESS in 6h 35m 00s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/12/20220522211149/NEC-ISCSI : SUCCESS in 6h 52m 43s","accounts_in_message":[],"_revision_number":12},{"id":"66c43c7366e764c66fe8783c00cbe40db49d9c3a","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-05-23 11:02:19.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"9172ddb5228b38973497f99d2cb45372737e1007","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-05-23 12:19:49.000000000","message":"Topic yadro-cinder-driver removed","accounts_in_message":[],"_revision_number":12},{"id":"2334737d4d1d8956890392448a96f054cdde3140","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-05-23 12:19:55.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"d503236df14db1061aba0cac697c41fc40b6a347","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-23 12:20:04.000000000","message":"Patch Set 13:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16055/","accounts_in_message":[],"_revision_number":13},{"id":"243a50f231ee19e5517af27ee35f3716b15643c6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-23 12:22:10.000000000","message":"Patch Set 13:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16055/ : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"a97202f1ebde925d3835d273caa91c274e6b1fca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-05-23 12:22:33.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"6371a5fe690ab9567a52afb4aa20751dec116d8c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-23 12:22:34.000000000","message":"Patch Set 13:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":13},{"id":"9a1c4da788dc8ab6fa640dd951ae395ece5a124f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-23 12:22:43.000000000","message":"Patch Set 14:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16056/","accounts_in_message":[],"_revision_number":14},{"id":"2d015f1e8efd6cd5d8fc9b3a12a930fb675417b4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-23 12:23:54.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":14},{"id":"3631426e2c4243e3b98157a468120dbaec893991","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-23 12:24:49.000000000","message":"Patch Set 14:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16056/ : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"c857642d9e53af076a62e8eedb5ce124a9fd4fe0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-23 12:55:42.000000000","message":"Patch Set 13:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1653308405589 : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"159e63f517efc29c0b145a436661ccf490b4ea17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 13:01:08.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/342f0d75b99c4b579e0696559f2be827 : SUCCESS in 20m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aa4f9650092641bc90e8c40d7c89c45d : FAILURE in 28m 28s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"4fb033052820bfd314d5ad6ae7d959b5e153a3e7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-23 13:04:21.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/42280c3008c846baac87a4ea9780b7b2 : RETRY_LIMIT in 11m 19s","accounts_in_message":[],"_revision_number":14},{"id":"4b06fe441fe7170b0a9cce74ec3beb60e23cc69e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-23 13:21:08.000000000","message":"Patch Set 14:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1653308560597 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"ae17891c032a4d045f94fc6d6d1bfcd60fca65e5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-23 14:21:09.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/14/iscsi/ : FAILURE in 1h 57m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/14/smb/ : SUCCESS in 1h 03m 02s","accounts_in_message":[],"_revision_number":14},{"id":"323d6ec37182b1e83f5600940494b4b2afea755c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 14:49:59.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44d499d23e0b44ce8dab676dacd2ee84 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2230824ebb344819bc1262663ea5f1e : SUCCESS in 11m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7329fe742cd64e0caca89b2e7fed8ba1 : SUCCESS in 11m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2e5287ad1644a5d9c2224dcbf5cd130 : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af84e22748c1428aac301e25fdd1aaed : SUCCESS in 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/2ba08abc977d48c0bd4897be70a942a0 : SUCCESS in 56m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c15e03cf5c2641f68c9c188c4f3e2060 : SUCCESS in 1h 13m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26253771eac4403caab7a2199d36e9dd : SUCCESS in 55m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/100175f6c2e842e9bb07c3594fd6f542 : SUCCESS in 7m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ca510d72d0df4f178f9b144e9d928c2f : SUCCESS in 13m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a552e8ebaf1f4ff28f00f6114998385b : SUCCESS in 6m 26s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2db6bcb5f05846b1a8ec2e1115f42f47 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c1b899b9eea14ed3aca20aeecd61109d : SUCCESS in 7m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a7443207296e4ce58888eff68e95efe6 : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/05d8ba31e12e484cbeda4630207907fe : SUCCESS in 45m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e612f4675f8b463f89674344606c52a3 : FAILURE in 6m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fa260ae2d77646e9acab8efa3b47c9de : SUCCESS in 1h 22m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aa3b948a56c145ba93a6487d84450825 : FAILURE in 1h 49m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c5a8900e55434890a4d1edd755ca7afd : SUCCESS in 1h 31m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9f8681bd45be4bc18e6a3aed90a27a71 : SUCCESS in 1h 12m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2b7ff900971445d39abaa41931aa6a9b : SUCCESS in 1h 30m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ce8c43e138c147bcab885b91a79e2ad3 : SUCCESS in 40m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5701419e22f14593a28053b68c5fd775 : SUCCESS in 1h 10m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0cf7203f91ab42ceb65d0579aaf478b2 : SUCCESS in 2h 24m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/178ceb6475e9425f9808e3e2bb71bdec : SUCCESS in 48m 07s","accounts_in_message":[],"_revision_number":14},{"id":"aa9a4bc99bd69f60d99ac85e70649373500bb3dc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-23 14:50:08.000000000","message":"Patch Set 14:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":14},{"id":"9f8f3c87cfe6b1f95c153446c346277446b680de","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-23 14:59:50.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/14/check/DellEMC_VNX_iSCSI/aad75a4/DellEMC_VNX_iSCSI/434 : FAILURE in 1s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/92/825492/14/check/DellEMC_VNX_FC/c8eb275/DellEMC_VNX_FC/453 : FAILURE in 9m 33s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":14},{"id":"d1339e0d244fd22e54b0ab363953e59e76b938b0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-23 15:08:40.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/825492/14/check/iscsi-driver-nimble/39174d4/39174d4.html : SUCCESS in 1h 35m 51s","accounts_in_message":[],"_revision_number":14},{"id":"d36b28a81ea1df4069eb5e2db5fd36aad76820d7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-23 15:28:21.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/14/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/7dbe144/ : TIMED_OUT in 3h 03m 53s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/825492/14/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/9cc6152/ : TIMED_OUT in 3h 03m 58s","accounts_in_message":[],"_revision_number":14},{"id":"54d69d24d279911ee0e6fe3cf0efac881aba56f7","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-05-23 15:44:08.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/14/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/68732b0/ : SUCCESS in 52m 12s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":14},{"id":"347646f70b65f93d8720bbaa069cf4486828a5e0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-23 16:07:57.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-13/ns5_nfs : SUCCESS in 3h 37m 27s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-13/ns5_iscsi : SUCCESS in 3h 44m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"88496acee9af5f898125196c90f9a5818458f4b2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-23 16:27:23.000000000","message":"Patch Set 14:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-14/ns5_nfs : SUCCESS in 3h 33m 36s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-14/ns5_iscsi : SUCCESS in 3h 41m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":14},{"id":"d2135108153f97283e8ef1944f637a0940ceeffb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-23 16:40:29.000000000","message":"Patch Set 14:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/14/check/tempest-dsvm-full-focal-py3/8ec5117/ : FAILURE in 2h 46m 34s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/14/check/devstack-plugin-ceph-dsvm-focal-py3/370ca5e/ : SUCCESS in 2h 56m 37s","accounts_in_message":[],"_revision_number":14},{"id":"8596ab7483b2968af4b90347d14e3cd7ed6c2ed2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-23 16:41:15.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/92/825492/14/check/DellEMC_PowerFlex/a1909df/DellEMC_PowerFlex/777 : SUCCESS in 1h 08m 37s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":14},{"id":"4e96bc49d5c5e8144735883b03bf2da2e0e52e76","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-05-23 16:50:30.000000000","message":"Patch Set 14:\n\nBuild failed.","accounts_in_message":[],"_revision_number":14},{"id":"8e3459351c411509aca4bf6de47b089374426ad9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-05-23 17:11:14.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/14/upstream-check/cinder-tempest-plugin-ontap-fcp/a3a4fe9/ : SUCCESS in 1h 14m 46s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/14/upstream-check/cinder-tempest-plugin-ontap-iscsi/3bc90e9/ : FAILURE in 2h 19m 28s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/14/upstream-check/cinder-tempest-plugin-ontap-nfs/d6e89ae/ : FAILURE in 1h 09m 54s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"2c8e737e41462a061e739da58dd6eab5ce115b46","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-23 17:52:17.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/14/20220524003502/NEC-V-FC : SUCCESS in 1h 44m 18s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/14/20220524003502/NEC-V-ISCSI : SUCCESS in 2h 17m 11s","accounts_in_message":[],"_revision_number":14},{"id":"8223cf6b127ff8250f9e5de3607707426d34720e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-05-23 18:10:19.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/14/check/DellEMC_PowerStore_iSCSI/d5ba014/DellEMC_PowerStore_iSCSI/522 : SUCCESS in 1h 19m 24s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/92/825492/14/check/DellEMC_PowerStore_FC/89df128/DellEMC_PowerStore_FC/547 : SUCCESS in 1h 11m 27s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":14},{"id":"52aad3888045f1c71cf9cb45efaa8321081cb12c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-05-23 19:12:34.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/92/825492/14/check/DellEMC_SC_FC/4e4b944/DellEMC_SC_FC/624 : SUCCESS in 45m 20s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/14/check/DellEMC_SC_iSCSI/504ad18/DellEMC_SC_iSCSI/645 : SUCCESS in 1h 32m 45s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":14},{"id":"34844d505d996ff0078df30a2435b16fc0831033","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-05-23 19:54:21.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/825492-14/PVME-iSCSI/ : SUCCESS in 1:07:32\n- PVME-FC https://os-logs.seagate.com/dell-ci/825492-14/PVME-FC/ : SUCCESS in 1:04:43","accounts_in_message":[],"_revision_number":14},{"id":"1f890ea0ff198d110badb687a36824b44b9d40c5","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-05-23 20:24:24.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/825492-14/Seagate-iSCSI/ : SUCCESS in 1:10:55\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/825492-14/Seagate-FC/ : SUCCESS in 1:17:49","accounts_in_message":[],"_revision_number":14},{"id":"422097f2de2febea84f7c92fc54552ffe22a2652","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-23 23:59:33.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/825492/14/check/DellEMC_Unity_FC/abce5fe/DellEMC_Unity_FC/815 : SUCCESS in 1h 09m 21s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/14/check/DellEMC_Unity_iSCSI/c4b3804/DellEMC_Unity_iSCSI/783 : SUCCESS in 1h 39m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"6f65f44758d7efbcf1a84543e56af1396341e4ef","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-24 04:57:50.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/825492-14/Lenovo-iSCSI/ : FAILURE in 3:22:37\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/825492-14/Lenovo-FC/ : SUCCESS in 6:01:58","accounts_in_message":[],"_revision_number":14},{"id":"bd44e01824ba5c26495d9d5cca20c257022ea5fc","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-05-24 11:52:04.000000000","message":"Patch Set 14: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":14},{"id":"d4110bf30316717c4d84e63ddf0591b54a1159e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-05-25 06:36:57.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"b95133ff7d78ce4406c1efa0d5ad386bcbf52fda","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-25 06:37:08.000000000","message":"Patch Set 15:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16067/","accounts_in_message":[],"_revision_number":15},{"id":"1e6ed6a99f758fd91683b494c9923eddcf131f0c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-25 06:38:45.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":15},{"id":"402714c52ce8462464a0f5cc66ba60b5b896084d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-25 06:39:29.000000000","message":"Patch Set 15:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16067/ : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"223d86ae231f01798632ad84a927746b2693de18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 07:03:10.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/63315ff87fd44ce3bb7975f4e312cfbb : SUCCESS in 19m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8efe63a450be4812be57c6d0c8a71d42 : SUCCESS in 23m 29s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"dd0b86f186636f9eb9ba6076078b19afa95c046b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-25 07:04:16.000000000","message":"Patch Set 15:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1653460625199 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"8780a90a4a74ed50e1318a75349bbdfd531277bf","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-25 07:52:39.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/15/iscsi/ : SUCCESS in 1h 15m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/15/smb/ : SUCCESS in 1h 10m 29s","accounts_in_message":[],"_revision_number":15},{"id":"e23389cf65261b4d9e5daf78f704ea681c3798d8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-25 08:01:55.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7f1236bd36034a04a251fb2f9d27fed1 : SUCCESS in 1h 23m 49s","accounts_in_message":[],"_revision_number":15},{"id":"b36dad80ba2103996565fe45e17fc663dd883b82","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-25 08:22:42.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/825492/15/check/iscsi-driver-nimble/3f3ef19/3f3ef19.html : SUCCESS in 1h 44m 36s","accounts_in_message":[],"_revision_number":15},{"id":"ea388fd4153c97937623f5dac253b7eed9b836e8","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-25 08:54:36.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/15/20220525153700/NEC-V-FC : SUCCESS in 1h 55m 31s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/15/20220525153700/NEC-V-ISCSI : SUCCESS in 2h 17m 32s","accounts_in_message":[],"_revision_number":15},{"id":"d5bc3cf9401039cf3386f13f0f2a95d5f41b60c8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-25 09:02:47.000000000","message":"Patch Set 15:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/15/check/tempest-dsvm-full-focal-py3/2c29bb0/ : SUCCESS in 1h 37m 46s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/15/check/devstack-plugin-ceph-dsvm-focal-py3/b24cf4e/ : SUCCESS in 2h 23m 50s","accounts_in_message":[],"_revision_number":15},{"id":"8cc1f10cb7c317fe56dcf7130cf03a252e2fa5f3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-05-25 09:08:48.000000000","message":"Patch Set 14:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/14/20220525111146/NEC-FC : FAILURE in 6h 52m 55s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/14/20220525111146/NEC-ISCSI : SUCCESS in 6h 56m 59s","accounts_in_message":[],"_revision_number":14},{"id":"ce15679f4991d72da4eb14f40450a24f45f94ca8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-25 09:41:39.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/15/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b76573e/ : TIMED_OUT in 3h 03m 08s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/825492/15/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/09d7634/ : FAILURE in 2h 43m 31s","accounts_in_message":[],"_revision_number":15},{"id":"5b1058f9ebadd7e683e9370f9837bb48ebf0925c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 09:55:03.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e9cd4792b22452b8a43ea4eb9c0450b : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bc5aa5d08f54104892124fee1e16ffa : SUCCESS in 11m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7380d1b19d444cdf8c1d8478591d9ea3 : SUCCESS in 11m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ab88d8933724930b1cf83f420fc690e : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/746e3eeb8e3e401f97fb6de280c81cf8 : SUCCESS in 12m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/19b4586698474bb8a0dbe6abc6500fed : SUCCESS in 55m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f13fa55e153411a988881ce2753f89a : SUCCESS in 1h 12m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e495e156279e49ecb296515757337719 : SUCCESS in 49m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/586de3e40d604e2d971532e2347c25f8 : SUCCESS in 8m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1b8b668cc90a4686a4fbdfec727dcdea : SUCCESS in 16m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e9aa55b97e614f84a42766b1e74fa0b8 : SUCCESS in 4m 41s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/edfe9ac1b92442609754fd8a25ada94b : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5109ac86ad0a4fb1b1950ba324359fed : SUCCESS in 6m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dea973206345480b8a7db1141d22da91 : SUCCESS in 5m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2c09c1ae382d45478a2150e0f582bff0 : SUCCESS in 37m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0e18402b08ec4c32973507a0bdb4435b : FAILURE in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4a8c2c75d5ff4cffba5d20dacdda7579 : SUCCESS in 1h 16m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f49a6a6322f04908b760a24818ca3f45 : TIMED_OUT in 3h 05m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/571cf8ea5c024d7fa3e81764b368e980 : SUCCESS in 1h 49m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/644e75395dcb49e0b9d7f07f275213e5 : SUCCESS in 59m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2b1bee027b0049d780fc681bae1ca4ba : SUCCESS in 1h 42m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/832c696ef5dd443ba58e8340aab06a07 : SUCCESS in 55m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3d8b0081e77a41879b5078acc6a615d2 : SUCCESS in 1h 12m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8868e0ea296342a69a07f0d442cf8196 : SUCCESS in 2h 40m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ed95f225f724e0eadb5a7897dc55e37 : SUCCESS in 48m 52s","accounts_in_message":[],"_revision_number":15},{"id":"3b80b6dd725879c200abb23b978c7730f774c5d3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-25 09:55:12.000000000","message":"Patch Set 15:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":15},{"id":"dc1f85bd315175c943cbd5fac02b6a1b22428375","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-25 10:05:37.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/15/check/DellEMC_VNX_iSCSI/831c6a7/DellEMC_VNX_iSCSI/448 : FAILURE in 1s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/92/825492/15/check/DellEMC_VNX_FC/b392ae8/DellEMC_VNX_FC/467 : FAILURE in 10m 14s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":15},{"id":"2fe16a1147f1fd1ad4d44d164c99c4aa5e11f2f4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-25 10:06:39.000000000","message":"Patch Set 15:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-15/ns5_nfs : SUCCESS in 3h 27m 56s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-15/ns5_iscsi : SUCCESS in 3h 29m 13s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":15},{"id":"ad8260b3f46572405e055ce00261fae553cd2cbb","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-05-25 10:07:31.000000000","message":"Patch Set 14:\n\n(5 comments)","accounts_in_message":[],"_revision_number":14},{"id":"eaa9b4124c637096a64da5b9caaeecec60a4c865","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-05-25 10:48:40.000000000","message":"Patch Set 15:\n\n(7 comments)","accounts_in_message":[],"_revision_number":15},{"id":"1914c0f78052a77455c76828ca244975d47def99","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-05-25 10:55:28.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/15/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/64f5b8e/ : SUCCESS in 59m 02s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":15},{"id":"b13dec2a79fe6e4a3c2588e9a84737edd7a08a38","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-25 11:05:28.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/92/825492/15/check/DellEMC_PowerFlex/6f64ca8/DellEMC_PowerFlex/791 : SUCCESS in 1h 09m 48s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"7a82ebf2e6ccd3d4d9e20704b74adf90e3fe8f20","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-25 11:17:16.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/825492/15/check/DellEMC_Unity_FC/af6540f/DellEMC_Unity_FC/834 : SUCCESS in 1h 10m 28s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/15/check/DellEMC_Unity_iSCSI/f9b931e/DellEMC_Unity_iSCSI/800 : ABORTED in 3h 00m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"4157e29f6d51a2d4606d595573b59869474e22fc","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-05-25 11:25:56.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/92/825492/15/check/DellEMC_SC_FC/d7a2e1d/DellEMC_SC_FC/638 : SUCCESS in 44m 41s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/15/check/DellEMC_SC_iSCSI/27be0bb/DellEMC_SC_iSCSI/659 : SUCCESS in 1h 30m 36s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":15},{"id":"47f950282af5f2e04ea775252c72872fa1915643","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-05-25 11:41:05.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/15/upstream-check/cinder-tempest-plugin-ontap-fcp/7baff50/ : SUCCESS in 1h 06m 36s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/15/upstream-check/cinder-tempest-plugin-ontap-iscsi/f67ea53/ : SUCCESS in 1h 43m 51s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/15/upstream-check/cinder-tempest-plugin-ontap-nfs/61a62ff/ : FAILURE in 1h 42m 34s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"64d739adaba7914be9b561f192241dc21b1c8da0","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-05-25 12:17:48.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/825492-15/HPMSA-iSCSI/ : SUCCESS in 54:56.14\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/825492-15/HPMSA-FC/ : SUCCESS in 52:27.28","accounts_in_message":[],"_revision_number":15},{"id":"e4abb11b7b045949cfdfccccf3f71f7b93718880","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-05-25 12:37:03.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/825492-15/PVME-iSCSI/ : SUCCESS in 1:04:46\n- PVME-FC https://os-logs.seagate.com/dell-ci/825492-15/PVME-FC/ : SUCCESS in 1:03:05","accounts_in_message":[],"_revision_number":15},{"id":"1929f461b885e69e36989302df744947ddf6f7c1","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-05-25 12:49:02.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/15/check/DellEMC_PowerStore_iSCSI/e7e0599/DellEMC_PowerStore_iSCSI/536 : FAILURE in 2h 53m 38s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/92/825492/15/check/DellEMC_PowerStore_FC/a2950a5/DellEMC_PowerStore_FC/561 : FAILURE in 2h 53m 35s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":15},{"id":"01371cd702092ff2ffe277f25b76c1eb9a554961","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-05-25 13:17:55.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/825492-15/Seagate-iSCSI/ : SUCCESS in 1:07:19\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/825492-15/Seagate-FC/ : SUCCESS in 1:02:32","accounts_in_message":[],"_revision_number":15},{"id":"bec918b022c8d0adbd067d3c5f778da07ccb0e81","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-26 06:13:55.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/825492-15/Lenovo-iSCSI/ : SUCCESS in 1:07:33\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/825492-15/Lenovo-FC/ : SUCCESS in 1:03:14","accounts_in_message":[],"_revision_number":15},{"id":"de470ba697274800be898955b6832076cb08d18f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-05-26 15:49:28.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"0aac9fdb7055ef41e5c2be0e7736e2f2ee2f0345","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-26 15:49:35.000000000","message":"Patch Set 16:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16082/","accounts_in_message":[],"_revision_number":16},{"id":"6dfb8d424c51a633d5d7f06b22f5fb9ebd16d1a1","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-05-26 15:50:55.000000000","message":"Patch Set 16:\n\n(5 comments)","accounts_in_message":[],"_revision_number":16},{"id":"a13cb5c88f5752a22512202f59e46000d2bab680","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-26 15:51:40.000000000","message":"Patch Set 16:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16082/ : FAILURE","accounts_in_message":[],"_revision_number":16},{"id":"6171a6e2c25e320dffbf708cc35a3ae8fc8c4aa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-05-26 15:53:30.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"9113264dc1ad8a00bd3480e4b9ffda7eee5c67dc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-26 15:53:38.000000000","message":"Patch Set 17:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16083/","accounts_in_message":[],"_revision_number":17},{"id":"8418a3743b8fb260798aecb46138529cc8a721e1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-26 15:55:39.000000000","message":"Patch Set 17:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16083/ : FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"e6bc0f46f26efdc21a816339f4696ee58230a60b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 18:19:39.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f2bb79c8d434d57ad5db85236aca3ba : SUCCESS in 23m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/240bd7d19e2a440d81880f8e950a4de5 : SUCCESS in 24m 35s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"a07ed66d050cca585cad3daca8919e531539cc41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 18:25:45.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc9cac5d77454889aff2626e791b3f01 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7f249b380334697a06a6afe6de9219f : SUCCESS in 12m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cceb3574dc6442faa5ed6ba8bfe1282 : SUCCESS in 11m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69f695302530410588c3e5bd0305b4a2 : SUCCESS in 10m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/897e0fa9fecb4e568ff587ceece0ecc9 : SUCCESS in 11m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/636e270d820046eda93e4ee9228c5720 : SUCCESS in 53m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/12deb2ab0dfb4d00a89786befaf9cc57 : SUCCESS in 1h 22m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d19905917cb46ae9935bba24347df15 : SUCCESS in 53m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5c51d6a2f20453da288d71858912aa5 : SUCCESS in 9m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1fdc980164074083b8e5a56693bde523 : SUCCESS in 15m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/69a45405f2434ea0a6d4d22d79c63468 : SUCCESS in 7m 53s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c0213c8673714985954b54467a7ce4a0 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2796ce98fbca403f917c18b1acf371f0 : SUCCESS in 8m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f9c234939aa0474bbc88b7758d37ed57 : SUCCESS in 7m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3718919197bd4ec98fdb5fc55e164293 : FAILURE in 19m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8356f537ef3a4977b5c1b7240208338f : FAILURE in 9m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/17c7686e93d34a5b93b70a46864ea8b1 : SUCCESS in 1h 17m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aa3be2201170470ca8d3a2880540b967 : FAILURE in 1h 24m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5a807c7260504e8c8392ac6312b3ce42 : SUCCESS in 1h 40m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5a30f644f889422e9b492c4b28a4086b : SUCCESS in 1h 09m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f6340cf1c72c4fd9a69b09f1ceb730a3 : SUCCESS in 1h 20m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3246370181184f508bfeba94bf5ca636 : SUCCESS in 40m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/174a3495b3c84d9eb92bdb970b885a7d : SUCCESS in 1h 24m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c131ead75cb1485da0b598f4028ea975 : SUCCESS in 2h 07m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8b65aa178fa404a9d278a202998efd6 : SUCCESS in 1h 01m 18s","accounts_in_message":[],"_revision_number":17},{"id":"68ef58a6106d2a7ff9a6412a63aa7f8ac485273c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-26 18:25:56.000000000","message":"Patch Set 17:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":17},{"id":"4ec328fbeee43a3cb95f80624bbc9247a3125804","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-26 19:16:54.000000000","message":"Patch Set 16:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1653580174673 : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"1f2744c7ed3370b3a7e90ab90e004e608cb5b884","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-26 19:20:57.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/17/check/DellEMC_VNX_iSCSI/a453b8c/DellEMC_VNX_iSCSI/466 : FAILURE in 1s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/92/825492/17/check/DellEMC_VNX_FC/095fb8b/DellEMC_VNX_FC/485 : FAILURE in 12m 19s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":17},{"id":"fac4b66f3c33752c3f32a5437bc9283c91036282","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-26 19:40:48.000000000","message":"Patch Set 17:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1653580419689 : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"432434faaeb37024f25ea3a3660587911ce2d686","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-26 19:55:30.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/17/iscsi/ : SUCCESS in 1h 41m 49s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/17/smb/ : SUCCESS in 1h 30m 37s","accounts_in_message":[],"_revision_number":17},{"id":"2259f980a0549f1fc28638b3ae6b9f0790345233","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-05-26 20:00:05.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-92-825492-17/1653589545 : SUCCESS in 18m 19s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":17},{"id":"6ff98cfa484673558d67be3ace54f58174fccfcc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-05-26 21:09:43.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/17/upstream-check/cinder-tempest-plugin-ontap-fcp/eae1e8e/ : SUCCESS in 1h 01m 25s\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/4f4b01be421d4bfda6c8d3ebb8c31bed : RETRY_LIMIT in 4m 30s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/17/upstream-check/cinder-tempest-plugin-ontap-nfs/c137b22/ : FAILURE in 2h 26m 33s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":17},{"id":"283c3d635c076494bcf4b6a591e3640fa5fb1f6e","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-05-26 21:51:04.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/17/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/dd14e09/ : SUCCESS in 1h 58m 59s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":17},{"id":"ddb48699fd5a3f3f71cadc79f3beaf82601c4b03","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-26 22:17:28.000000000","message":"Patch Set 16:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-16/ns5_nfs : SUCCESS in 3h 39m 14s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-16/ns5_iscsi : SUCCESS in 3h 45m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":16},{"id":"841c44517a9b7be212f455c5da041a4d87fcde62","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-26 22:38:00.000000000","message":"Patch Set 17:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-17/ns5_nfs : SUCCESS in 3h 38m 0s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-17/ns5_iscsi : SUCCESS in 3h 43m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":17},{"id":"451469bd034436376350552ebcc814db41fe8155","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-26 22:49:50.000000000","message":"Patch Set 17:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/17/check/tempest-dsvm-full-focal-py3/74a03cc/ : SUCCESS in 2h 22m 41s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/17/check/devstack-plugin-ceph-dsvm-focal-py3/1fe0d59/ : SUCCESS in 2h 41m 44s","accounts_in_message":[],"_revision_number":17},{"id":"f9e9cec1519cfcd73c2b65df7e04401d2478357e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-27 00:19:36.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/825492/17/check/iscsi-driver-nimble/d7d0037/d7d0037.html : FAILURE in 59m 42s","accounts_in_message":[],"_revision_number":17},{"id":"86cf7fcf0e97d24a44659ca5b9947fbb73f886e2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-27 01:46:25.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/92/825492/17/check/dsvm-tempest-cisco-zonemanager-job/2b8dc81/ : SUCCESS in 1h 10m 47s","accounts_in_message":[],"_revision_number":17},{"id":"ff262e379a04ebe1f67150d3d87a5a6ebcfb5500","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-27 02:24:56.000000000","message":"Patch Set 17:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/92/825492/17/check/DellEMC_PowerFlex/7f19523/DellEMC_PowerFlex/806 : SUCCESS in 1h 17m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":17},{"id":"3ccb47c9bca7bd20657a54eb9938ef337301bb48","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-05-27 02:55:54.000000000","message":"Patch Set 17:\n\nBuild failed.","accounts_in_message":[],"_revision_number":17},{"id":"96daf8e9498564890ba604b0aa43750b45e4f3f1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-27 03:09:00.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/17/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f375dcd/ : TIMED_OUT in 3h 03m 22s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/825492/17/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/43ce795/ : SUCCESS in 1h 18m 22s","accounts_in_message":[],"_revision_number":17},{"id":"42732d7ca42004a5c503e10c4a45f230bca5eb50","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-05-27 08:22:11.000000000","message":"Patch Set 17:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/17/check/DellEMC_PowerStore_iSCSI/f3a58c1/DellEMC_PowerStore_iSCSI/551 : SUCCESS in 1h 27m 55s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/92/825492/17/check/DellEMC_PowerStore_FC/68cd82b/DellEMC_PowerStore_FC/576 : SUCCESS in 1h 14m 31s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":17},{"id":"8427ec66099263e4546143941d9e8a24790a73cf","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-05-27 08:28:17.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"6871bcffc085d78e6faaffc7d71c1fba4ceb375d","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-05-27 08:33:42.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"0dffe60b073222c4666bce44b2b42c17a210a09e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-05-27 08:34:04.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"2fa84f182124fd2360bda792c4e21c827f1d7ea3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-27 08:34:16.000000000","message":"Patch Set 18:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16087/","accounts_in_message":[],"_revision_number":18},{"id":"fcfead6071300c91a76e86a908c96d87557ba6fc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-27 08:36:14.000000000","message":"Patch Set 18:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16087/ : FAILURE","accounts_in_message":[],"_revision_number":18},{"id":"6bdd99e3ba53a8606bb3f5a907a48e08d3443fe3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 08:56:52.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f9b442833d24b3b9a94b2922c18765e : SUCCESS in 18m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/367b4eb2f6fe4935ab2f008ed9881914 : SUCCESS in 20m 07s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"0c1a0023f7620f74b1e4055c24988d9bfc2d7fba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-27 09:00:43.000000000","message":"Patch Set 18:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1653640452127 : SUCCESS","accounts_in_message":[],"_revision_number":18},{"id":"4f6530ed40777a05668712fe15c2f883507c93f9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-27 09:36:49.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/825492/18/check/iscsi-driver-nimble/2e779f6/2e779f6.html : FAILURE in 1h 01m 39s","accounts_in_message":[],"_revision_number":18},{"id":"ff3466ece81eecefe97eb2ffa22c10e4ec101fc4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-27 10:09:43.000000000","message":"Patch Set 18:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/18/iscsi/ : FAILURE in 1h 35m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/18/smb/ : SUCCESS in 1h 19m 47s","accounts_in_message":[],"_revision_number":18},{"id":"2a061dcaad482e68115fc2a65d37974e0e71dc1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 10:54:50.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf50807170c246d1a0b8f9ef8f4cc89e : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0287f0b2dc84310bffd75a6fe89fc75 : SUCCESS in 9m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/447b5bdfe3b5492b9a7dd0fb6f753147 : SUCCESS in 12m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ef331fe7cda4ae59c7fbe2c7293c95c : SUCCESS in 11m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b9118dac5cd4c508776d8641fcd26b7 : SUCCESS in 12m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/2cd1e488c2664aed821e33b25b75268d : SUCCESS in 53m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1752d7c8dd454011ae5ee8485fcc0342 : SUCCESS in 1h 20m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de507d53dbca4e7c984a84e28ee195e3 : SUCCESS in 47m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88043fbdf9644b209bf1b204206fdd07 : SUCCESS in 8m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0203251dc0054dab9eebb241ea6846da : SUCCESS in 13m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7c0717112db2402392dc85f5652da183 : SUCCESS in 6m 50s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dff703ec5d7b4dd2b33c1ad88d0f3b9f : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8d00e729e15143a2909dd70850b1ea72 : SUCCESS in 8m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6adf96075e304f02b6578b4aa0538adc : SUCCESS in 7m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b36157dc1bcf4cc7ab667d3e3459f679 : SUCCESS in 39m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2395a65d66b04235b77a2ec70e0914a4 : FAILURE in 8m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a04c7b7810004bfb91bb0ad99d19128b : SUCCESS in 1h 12m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0bac83a907d74cd1928cb874ab885afb : FAILURE in 1h 24m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9f91382e31174ad8908859cb4cbf17ec : SUCCESS in 1h 45m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6555961a4fcc4f9094ef2ee247611530 : SUCCESS in 1h 02m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8ccbd5dd89b6450eb9fc451575e76551 : SUCCESS in 1h 17m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/883196f1436d4f2ba8aa1806be6c6058 : SUCCESS in 42m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/058ba7856ebb4058b5ec651da2f24650 : SUCCESS in 1h 13m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3b5e8dd22b234199a5c5c4bf825e8e99 : SUCCESS in 2h 18m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0e44dc35109481d94b0d58b14836b2d : SUCCESS in 45m 47s","accounts_in_message":[],"_revision_number":18},{"id":"c036b2b6ecd23cb6410f53ec6349ed6f65535534","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-27 10:54:57.000000000","message":"Patch Set 18:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":18},{"id":"6d2f55eea44be0dda259b601bc9d2c1b7ff2b291","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-27 10:55:26.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/825492/18/check/DellEMC_Unity_FC/bf56985/DellEMC_Unity_FC/869 : FAILURE in 5s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/18/check/DellEMC_Unity_iSCSI/43a5559/DellEMC_Unity_iSCSI/814 : FAILURE in 9s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":18},{"id":"45c5ddaa363b692ae281e9d1e6b5b899a97eacda","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-27 11:39:19.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/18/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/31fb899/ : TIMED_OUT in 3h 03m 12s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/825492/18/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/865018f/ : SUCCESS in 1h 03m 26s","accounts_in_message":[],"_revision_number":18},{"id":"e599fe65e9e398d138e3b3837c30d4afd4d8c957","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-27 11:59:39.000000000","message":"Patch Set 18:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/18/check/tempest-dsvm-full-focal-py3/6f97a1c/ : SUCCESS in 2h 06m 51s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/18/check/devstack-plugin-ceph-dsvm-focal-py3/3013393/ : SUCCESS in 2h 44m 44s","accounts_in_message":[],"_revision_number":18},{"id":"79e599d9fb4da7e32682a7399936133e5d58a288","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-27 12:06:23.000000000","message":"Patch Set 18:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-18/ns5_nfs : SUCCESS in 3h 31m 51s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-18/ns5_iscsi : SUCCESS in 3h 30m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":18},{"id":"9166a276b53d7603ce319d899820e424e5daca75","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-05-27 12:25:32.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/18/upstream-check/cinder-tempest-plugin-ontap-fcp/a71f12b/ : SUCCESS in 1h 03m 34s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/18/upstream-check/cinder-tempest-plugin-ontap-iscsi/9b30911/ : FAILURE in 1h 22m 21s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/18/upstream-check/cinder-tempest-plugin-ontap-nfs/e29d38b/ : FAILURE in 1h 15m 08s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":18},{"id":"d7cfaca4f53dcb4dbcb20d86822c012de5d84e10","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-05-27 12:37:50.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/18/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/299f409/ : FAILURE in 1h 35m 00s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":18},{"id":"bb215479043b40a9c0abe0fbed9d6bc504984184","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-27 16:01:15.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/92/825492/18/check/dsvm-tempest-cisco-zonemanager-job/d6cbcb7/ : SUCCESS in 1h 08m 09s","accounts_in_message":[],"_revision_number":18},{"id":"7fed6f94673ef52b08934c0a4d775011600163d5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-28 10:06:50.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4f34389f9c8c42e9a317c51cdacbb527 : FAILURE in 48m 16s","accounts_in_message":[],"_revision_number":18},{"id":"5348ec003abf624841deb4b283d8faca7085e071","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-28 19:21:05.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/18/20220529021408/NEC-V-FC : SUCCESS in 1h 47m 40s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/18/20220529021408/NEC-V-ISCSI : SUCCESS in 2h 06m 54s","accounts_in_message":[],"_revision_number":18},{"id":"c13b01bb76e73d0107e8ec8c6cf5f74bea3530e7","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2022-05-31 04:50:41.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"e653a0b024399eaf5a408fb6f90d590b2caabbbc","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-05-31 06:13:18.000000000","message":"Patch Set 18:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/18/check/DellEMC_PowerStore_iSCSI/21e0bf6/DellEMC_PowerStore_iSCSI/553 : SUCCESS in 1h 22m 03s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/92/825492/18/check/DellEMC_PowerStore_FC/94df937/DellEMC_PowerStore_FC/578 : SUCCESS in 1h 21m 20s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":18},{"id":"c560ced47aa44cb713e92cc1caceb6494efe7a5d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-01 13:41:02.000000000","message":"Patch Set 10:\n\nFAILED","accounts_in_message":[],"_revision_number":10},{"id":"ff074ac70407987c9b987df98b4380afeab0c32c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-01 13:41:49.000000000","message":"Patch Set 10:\n\nFAILED","accounts_in_message":[],"_revision_number":10},{"id":"52959465f55a784e3d3fb8b6e5473e46f6549e72","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-01 13:48:12.000000000","message":"Patch Set 10:\n\nFAILED","accounts_in_message":[],"_revision_number":10},{"id":"18179398a363422104aeae6d7bc9530c3de2748e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-01 13:51:08.000000000","message":"Patch Set 10:\n\nFAILED","accounts_in_message":[],"_revision_number":10},{"id":"e6dc4fdab359bbab6d1bc7cba98dc51a75fce795","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-01 13:52:13.000000000","message":"Patch Set 10:\n\nFAILED","accounts_in_message":[],"_revision_number":10},{"id":"ad9ed4c9cf18312a6cd5e85eece1e723afd2a526","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-01 13:53:19.000000000","message":"Patch Set 10:\n\nFAILED","accounts_in_message":[],"_revision_number":10},{"id":"2e151f219fff78d854a9865b24d51310e05d18d5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-01 13:55:35.000000000","message":"Patch Set 10:\n\nFAILED","accounts_in_message":[],"_revision_number":10},{"id":"26453628c997cd5a0b46ea7562b44e68a6a83087","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-01 14:00:27.000000000","message":"Patch Set 10:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-test/57/ : ABORTED\n\nnull","accounts_in_message":[],"_revision_number":10},{"id":"6834e63547dfc574de759ad5c38f6c84a3f26c5c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-01 14:02:44.000000000","message":"Patch Set 10:\n\nFAILED","accounts_in_message":[],"_revision_number":10},{"id":"bc9448280ae722d90e58f228dcc1856bef638d46","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-01 14:25:20.000000000","message":"Patch Set 10:\n\nFAILED","accounts_in_message":[],"_revision_number":10},{"id":"e714ca1e0806e063da89b9a661205e0b5ae553ec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-01 14:54:35.000000000","message":"Patch Set 10:\n\nFAILED","accounts_in_message":[],"_revision_number":10},{"id":"040b91742ef46db69762ab3a3506f834b09001bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-06-02 08:18:52.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"9c1d8aa7906a9fd9e36c31db5116960abf0c4d27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 08:47:12.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fecdea708db84ee88483ccd0c7816847 : SUCCESS in 12m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/14be310ca6774b60b578d53696a4f266 : SUCCESS in 25m 59s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"c118773b4916f36e11ff493f8627f597d1adcc4e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-02 08:58:15.000000000","message":"Patch Set 19:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1654157941585 : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"ec67e3ad2b9bb6887643839b28996d37a7be83fc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-02 09:30:06.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/19/iscsi/ : SUCCESS in 1h 10m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/19/smb/ : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":19},{"id":"04070a37c82af23557ad683e3644786796ee65b5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-02 09:41:40.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/92/825492/19/check/dsvm-tempest-cisco-zonemanager-job/37c6b3b/ : SUCCESS in 1h 12m 36s","accounts_in_message":[],"_revision_number":19},{"id":"a226d0c4b6186cb53ec8a27e5505bc64aedad6a6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-02 09:52:13.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/825492/19/check/iscsi-driver-nimble/5b0c5f5/5b0c5f5.html : FAILURE in 1h 32m 08s","accounts_in_message":[],"_revision_number":19},{"id":"7a77f2f5ac522aac8cbd21db3b8363cf85c7db8a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-02 09:57:08.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/825492/19/check/DellEMC_Unity_FC/9993ce7/DellEMC_Unity_FC/895 : SUCCESS in 1h 07m 54s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/19/check/DellEMC_Unity_iSCSI/97c683a/DellEMC_Unity_iSCSI/840 : SUCCESS in 1h 37m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":19},{"id":"fc1cb89eab8562ba8790f8e5edcd7a74815b57b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 10:34:16.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d281a400b52408bb746ac6712008f63 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4be21390cad5426a81ba4e14521edc2e : SUCCESS in 10m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67b282a65894477bbfe926678f477bbb : SUCCESS in 10m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eccc424bada347c396f73515b8097d08 : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eea7a44ba414ca5beca8bab427ca68d : SUCCESS in 10m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/782aa835363f42c1bc9bf2a416ac8d46 : SUCCESS in 50m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c75a409d53bb493eb91424102092ceb9 : SUCCESS in 1h 06m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ba3e856cffc4fc8a22cb1e5667b97fe : SUCCESS in 47m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd82d7e51bfd45dbab3bd82f4a8370c9 : SUCCESS in 8m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ca1c3b3a6d8a412ba8015b73d67314f0 : SUCCESS in 13m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2e665834a28a437a9f8267941b94aee0 : SUCCESS in 5m 29s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/760caca35e3a4f4eb471c7371375247a : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ac00a93af741463fa414af24032a4f2d : SUCCESS in 6m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/be27aab29b8544d78ff7d92568783b31 : SUCCESS in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/13dca45b47e14f77aedba8a78dea3569 : SUCCESS in 38m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c6cfa091fcd54650a10783aef92582c6 : FAILURE in 7m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0d807f3531b543d1a9c251a51febc5c1 : SUCCESS in 1h 15m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9cd3239d9b214029827fb1bfd7bcf557 : FAILURE in 1h 26m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f556407f5d0e45a2ad59027ef1457d90 : SUCCESS in 1h 42m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/60a5df6a26654796ab45f4653f5e6a30 : SUCCESS in 1h 00m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/68a51e32fa274f69b90a762dc96deb72 : SUCCESS in 1h 08m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/990a48f8a5414082be100706a02fffdb : SUCCESS in 40m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c946ce0ca2034f73beed779ba7334422 : SUCCESS in 1h 12m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/df829bc528194d19ae3073d17c8550b2 : SUCCESS in 2h 12m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5371e120ede9496d838eaeea05d2988f : SUCCESS in 45m 31s","accounts_in_message":[],"_revision_number":19},{"id":"05d395f9bf985054497b733703b7e004fdd9630f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-06-02 10:34:23.000000000","message":"Patch Set 19:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":19},{"id":"638022959c16916bd182f579c0948d63dacdc588","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-06-02 10:35:14.000000000","message":"Patch Set 19:\n\nBuild failed.","accounts_in_message":[],"_revision_number":19},{"id":"6d77fee082e7c2441cea78216239c0c060830cb1","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-06-02 10:35:42.000000000","message":"Patch Set 19:\n\nBuild failed.","accounts_in_message":[],"_revision_number":19},{"id":"1dad6299966ab0d39ff2f4a2cad0df6c444e8b2e","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-06-02 10:35:54.000000000","message":"Patch Set 19:\n\nBuild failed.","accounts_in_message":[],"_revision_number":19},{"id":"bf3e265397cd5f198e849e866a7cff51f526f383","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-06-02 10:36:15.000000000","message":"Patch Set 19:\n\nBuild failed.","accounts_in_message":[],"_revision_number":19},{"id":"265679ad88370a6a25c8a9f18bddd200b3b02290","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-02 10:41:09.000000000","message":"Patch Set 19:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/19/check/tempest-dsvm-full-focal-py3/40f5481/ : SUCCESS in 1h 39m 11s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/19/check/devstack-plugin-ceph-dsvm-focal-py3/37bd858/ : FAILURE in 2h 19m 51s","accounts_in_message":[],"_revision_number":19},{"id":"c4026ac4d43a7e5b6abd7a16775622a910b6cbb0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-02 11:23:51.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/19/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/aa5a518/ : TIMED_OUT in 3h 03m 10s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/825492/19/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/96fc884/ : FAILURE in 1h 01m 57s","accounts_in_message":[],"_revision_number":19},{"id":"f2fd985f973ba983d57f4eadef7d38c690e79620","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-06-02 11:44:09.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/92/825492/19/check/DellEMC_PowerFlex/4c69ea7/DellEMC_PowerFlex/818 : FAILURE in 1h 09m 32s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":19},{"id":"6ee745cf82792874ab0e114085c3954b298d4a13","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-02 12:00:59.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/19/upstream-check/cinder-tempest-plugin-ontap-fcp/3d03a81/ : SUCCESS in 1h 07m 15s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/19/upstream-check/cinder-tempest-plugin-ontap-iscsi/35c515a/ : FAILURE in 1h 25m 36s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/5d793c8953b54107b5af4caf1f8ef57d : RETRY_LIMIT in 3m 59s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":19},{"id":"7181de9028ad882fc2c1096f9de9e142a915b75b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-02 12:01:51.000000000","message":"Patch Set 19:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-19/ns5_nfs : FAILURE in 3h 29m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-19/ns5_iscsi : FAILURE in 3h 34m 28s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":19},{"id":"955bacac454d007093a8e0d5b56705513a67cc49","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-06-02 12:12:22.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/92/825492/19/check/DellEMC_SC_FC/3ded33d/DellEMC_SC_FC/646 : SUCCESS in 52m 09s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/19/check/DellEMC_SC_iSCSI/90ff3ff/DellEMC_SC_iSCSI/667 : FAILURE in 1h 37m 43s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":19},{"id":"ddb1a0274eedf7feb31ff5b812e4d019398191ac","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-06-02 12:19:48.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/19/check/DellEMC_PowerStore_iSCSI/a86a890/DellEMC_PowerStore_iSCSI/555 : SUCCESS in 1h 45m 12s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/92/825492/19/check/DellEMC_PowerStore_FC/1086a71/DellEMC_PowerStore_FC/580 : SUCCESS in 1h 44m 40s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":19},{"id":"a564abbef4d10c57d80425c3acefab3cda073409","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-06-02 12:35:28.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/19/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/4285f49/ : SUCCESS in 1h 52m 54s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":19},{"id":"ad5cc502a1d0e6aac98e9f367c99fd80c08655ee","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-02 12:58:09.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/825492/19/check/DellEMC_Unity_FC/4d744cd/DellEMC_Unity_FC/897 : SUCCESS in 1h 03m 11s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/19/check/DellEMC_Unity_iSCSI/47c3ffc/DellEMC_Unity_iSCSI/842 : SUCCESS in 1h 30m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":19},{"id":"b229e6dbdd911b3aa940a8f4e8fdde9f85d0c329","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-06-02 14:19:26.000000000","message":"Patch Set 19:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/19/20220602171855/NEC-V-FC : FAILURE in 5h 38m 35s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/19/20220602171855/NEC-V-ISCSI : FAILURE in 6h 00m 27s","accounts_in_message":[],"_revision_number":19},{"id":"51b7b9d577b6eb3b787a77817db65d2c5063595e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-03 11:20:28.000000000","message":"Patch Set 10:\n\nSUCCESSFUL","accounts_in_message":[],"_revision_number":10},{"id":"f83c2784d6ddba028d15384a029474346089d592","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-03 11:24:03.000000000","message":"Patch Set 10:\n\nFAILED \u003cSTARTED_STATS\u003e","accounts_in_message":[],"_revision_number":10},{"id":"c7d6325727ae54785e4c59e2010aac12c73eea3f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-03 11:28:15.000000000","message":"Patch Set 10:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-test/71/ : FAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/71 \n\nnull","accounts_in_message":[],"_revision_number":10},{"id":"94c953d8c3f9499d97800eccf12682ff6ddd811e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-03 11:37:11.000000000","message":"Patch Set 10:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/72","accounts_in_message":[],"_revision_number":10},{"id":"a3b9de385f6b1488f711d304f25f85f74ba4f00f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-03 14:04:18.000000000","message":"Patch Set 10:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-test/73","accounts_in_message":[],"_revision_number":10},{"id":"e3130c5946ca58ba5fa8b10c4f98221949a3d829","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-03 15:45:36.000000000","message":"Patch Set 10:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-test/74/ : ABORTED\n\nnull","accounts_in_message":[],"_revision_number":10},{"id":"46d75aceb37a65b4d31718274d2262d74f7afbc8","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-06-03 15:57:41.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"4056c65d739bfd18d80f96b4ef07b0ef9c15f9a2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-03 16:14:14.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"3d4ef66b075188f29797d6e4f6aa3a22f62c6039","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-03 16:22:08.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"61821750e9630eec187d652981431e876f2cd2cb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-03 18:42:15.000000000","message":"Patch Set 19:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-test/75","accounts_in_message":[],"_revision_number":19},{"id":"25691173f3b573142f2fb4fa2211466d9c019b29","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-06-06 06:54:16.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"2ad35addf1742e2489dd7fd5e00f4b02d7b130d8","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-06-06 10:53:11.000000000","message":"Patch Set 19: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"fb376cff14baff460245533f547c8669c81ca56b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-06-06 11:20:10.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"ca870e04d16157064f723191faced38581fba195","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-06-06 11:22:05.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"23d35fd966b64582bf3f3fe2474badeb529f2f12","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-06 11:39:19.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/df562b1056e74fe9960da56148dfe5e7 : RETRY_LIMIT in 5m 39s","accounts_in_message":[],"_revision_number":20},{"id":"842ea68c4426d97bf8a08f37d6d46a9a109a50c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 11:43:24.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/343cb553bbfc46a2ad412e25d7580e5f : SUCCESS in 18m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3235c90da4544df823ee433e3b81dfa : SUCCESS in 20m 19s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"57d8b053fd35ff4fd402b4a9986da0cc91379160","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-06 11:48:35.000000000","message":"Patch Set 20:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1654514416589 : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"fd5430acedcea561a3f5fc81087c65060afbe326","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-06 12:21:47.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/20/iscsi/ : SUCCESS in 1h 01m 06s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/20/smb/ : SUCCESS in 52m 15s","accounts_in_message":[],"_revision_number":20},{"id":"1926779ab41cfc99ebbee0337b11d1a9c8c69d6e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-06 13:03:38.000000000","message":"Patch Set 20:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/825492/20/check/DellEMC_Unity_FC/389d996/DellEMC_Unity_FC/921 : SUCCESS in 1h 07m 29s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/20/check/DellEMC_Unity_iSCSI/27bf12c/DellEMC_Unity_iSCSI/865 : SUCCESS in 1h 34m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":20},{"id":"386922ab6a6b033177a5f597d18d9822a4ece580","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-06-06 13:11:18.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/825492 : SUCCESS in 1h 36m 58s","accounts_in_message":[],"_revision_number":20},{"id":"a84d75e82b369ec06db8f9916b389025cd324947","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 13:23:39.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96818525da5241068d178a3b1a8ea0c8 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05493fe6470e40ab9b4e8a3320d0a421 : SUCCESS in 9m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51e99c34143042b7afc746343a981379 : SUCCESS in 9m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/228aae2898aa4195af66699733c19217 : SUCCESS in 9m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caa6a495703542d8a009a6c1178e4f4a : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/3d4d2055439f4ed7aacb24bf6a717555 : SUCCESS in 39m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f7d05801b38f4226ae8d9af6f757fe39 : SUCCESS in 1h 11m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfc20a1140f547b5b6c9a7e0f7f45119 : SUCCESS in 46m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/563e7fd43b6c4f6d8c7957eed10ed859 : SUCCESS in 7m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8c344c51e71a413ca16529a23a18dca3 : SUCCESS in 13m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8f3d80a1df5c45d387ebe20a00c0dda7 : SUCCESS in 6m 16s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bd624f49db024cde86299d13c49bec95 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ddaed857078d4eb7bcaa8b9ce4548faf : SUCCESS in 5m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4c3623915c7c4a6dac8894b5e3aea526 : SUCCESS in 6m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c972f24896df4e04b3a674b7f03fbad7 : SUCCESS in 39m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/70b71a96a46444d89e7ef89552604c3a : FAILURE in 7m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d5d7ee8f793847aab38e15e6042a01dc : SUCCESS in 1h 09m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ce90d6ec969b4ba7a872db574092e257 : FAILURE in 1h 32m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c2f19d6ac6dd43f7840b9addb0e4112e : SUCCESS in 1h 45m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0ca2e4ebe46c4cbf9ad7c95e6fc1056a : SUCCESS in 1h 03m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f22bc227c5514787afe2282507598653 : SUCCESS in 1h 07m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/252be84fbfce43b3a3e18ab711766fba : SUCCESS in 49m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/25ff4debf8234d94b9496609094a94a7 : SUCCESS in 1h 25m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d941afbf61c54dfda63e7db4a3dfbc59 : SUCCESS in 1h 59m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25159546335943319ed258d222ad2ac1 : SUCCESS in 48m 20s","accounts_in_message":[],"_revision_number":20},{"id":"6a549aa7670cf0a7c543b04c26ec5f64431a57bd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-06-06 13:23:48.000000000","message":"Patch Set 20:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":20},{"id":"91bd386b9a9bdcf7b7fc7ccd44ad56c037766e52","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-06-06 13:24:35.000000000","message":"Patch Set 20:\n\nBuild failed.","accounts_in_message":[],"_revision_number":20},{"id":"33e8e88bb804229d50638c485dfc23ceea2ca99b","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-06-06 13:25:02.000000000","message":"Patch Set 20:\n\nBuild failed.","accounts_in_message":[],"_revision_number":20},{"id":"2b33e751d552040feedfba50240b2b101c0bcc6d","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-06-06 13:25:06.000000000","message":"Patch Set 20:\n\nBuild failed.","accounts_in_message":[],"_revision_number":20},{"id":"8527979cc0e6657d0709c444917e88badb839f9c","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-06-06 13:25:36.000000000","message":"Patch Set 20:\n\nBuild failed.","accounts_in_message":[],"_revision_number":20},{"id":"8e5ef569a40354850e430a56067c7a8fa88df880","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-06 13:33:54.000000000","message":"Patch Set 20:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/20/check/tempest-dsvm-full-focal-py3/b327963/ : SUCCESS in 1h 39m 14s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/20/check/devstack-plugin-ceph-dsvm-focal-py3/dcbc4fc/ : SUCCESS in 2h 10m 38s","accounts_in_message":[],"_revision_number":20},{"id":"f1c45b959d7f040b6f67782167e6a35b61598d19","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-06 13:40:25.000000000","message":"Patch Set 20:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-test/76","accounts_in_message":[],"_revision_number":20},{"id":"92dc84412c2add1a3a87c8d755c614dab1f1ef53","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-06-06 13:42:06.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-92-825492-20/1654521819 : SUCCESS in 18m 24s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":20},{"id":"3745a3a6db311aa5000aee411e908e62ae84cd02","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-06 14:00:00.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/92/825492/20/check/dsvm-tempest-cisco-zonemanager-job/c0131c2/ : SUCCESS in 1h 11m 26s","accounts_in_message":[],"_revision_number":20},{"id":"1b6abc98169eebf67830d51aaac710077bd5e806","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-06 14:04:50.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/825492/20/check/iscsi-driver-nimble/829afe1/829afe1.html : FAILURE in 1h 33m 31s","accounts_in_message":[],"_revision_number":20},{"id":"5885b75aa2bbe3adc8c59a5e63a1ac4ca66d577d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-06 14:25:36.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/20/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/be81d71/ : TIMED_OUT in 3h 03m 24s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/825492/20/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b0721af/ : FAILURE in 1h 21m 30s","accounts_in_message":[],"_revision_number":20},{"id":"fedbe211e9d1d8edd79d7e345115a82c1f3062d9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-06-06 14:29:27.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/92/825492/20/check/DellEMC_PowerFlex/24a94e2/DellEMC_PowerFlex/825 : FAILURE in 1h 05m 28s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":20},{"id":"8bbea93245ac752d865cb005ff483dff2e2948bd","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-06-06 14:29:32.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/20/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/099614f/ : SUCCESS in 1h 03m 38s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":20},{"id":"dd2ce4b500f7a8ce9159d54e0bb408e229b9c828","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-06 14:32:22.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/20/upstream-check/cinder-tempest-plugin-ontap-fcp/0f698b3/ : SUCCESS in 1h 01m 33s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/20/upstream-check/cinder-tempest-plugin-ontap-iscsi/3f835ec/ : FAILURE in 1h 06m 28s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/20/upstream-check/cinder-tempest-plugin-ontap-nfs/f8e8288/ : FAILURE in 50m 37s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":20},{"id":"880db7cc68f41587cdec3b320855e9a39ce7a372","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-06-06 14:53:17.000000000","message":"Patch Set 20:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/20/check/DellEMC_PowerStore_iSCSI/196564d/DellEMC_PowerStore_iSCSI/562 : SUCCESS in 1h 29m 17s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/92/825492/20/check/DellEMC_PowerStore_FC/0279600/DellEMC_PowerStore_FC/587 : SUCCESS in 1h 17m 21s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":20},{"id":"0b3421e420259c51d63e6e03cdf4933a76eaa7f9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-06 14:57:28.000000000","message":"Patch Set 20:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-20/ns5_nfs : FAILURE in 3h 31m 29s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-20/ns5_iscsi : FAILURE in 3h 36m 47s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":20},{"id":"47008ce513db6282f1e2256c6987cad5a47ec2c2","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-06-06 14:59:24.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/92/825492/20/check/DellEMC_SC_FC/e6fb962/DellEMC_SC_FC/653 : SUCCESS in 44m 45s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/20/check/DellEMC_SC_iSCSI/eb065c5/DellEMC_SC_iSCSI/674 : FAILURE in 1h 35m 26s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":20},{"id":"7bb1132d0768ec448ed0ac215349edef7f6a5d57","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-06 16:17:57.000000000","message":"Patch Set 20:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/825492/20/check/DellEMC_Unity_FC/e14ee8f/DellEMC_Unity_FC/923 : SUCCESS in 1h 13m 09s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/20/check/DellEMC_Unity_iSCSI/84bbfd6/DellEMC_Unity_iSCSI/867 : SUCCESS in 1h 43m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":20},{"id":"92517e736a8b2102d0ccc506b0d7f6fbaf9ec007","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-06-06 21:57:38.000000000","message":"Patch Set 20:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/20/20220607005937/NEC-V-FC : FAILURE in 5h 12m 03s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/20/20220607005937/NEC-V-ISCSI : FAILURE in 5h 57m 58s","accounts_in_message":[],"_revision_number":20},{"id":"147f536131b696d2aa57c3e8258ee6eea26fde7c","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-06-07 06:07:02.000000000","message":"Patch Set 20: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"040536ddf5b85eee1fb07fb44eb6a7b29649b435","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-06-08 14:49:30.000000000","message":"Patch Set 20: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":20},{"id":"af7a75be65321017a71e050efc3f3e455572e6dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-06-09 13:04:23.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"fc897fba43c99f9f1e58eebde5194c49fca9b91c","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-06-09 13:04:57.000000000","message":"Patch Set 21:\n\n(12 comments)","accounts_in_message":[],"_revision_number":21},{"id":"49abb46398c6dd93c0187d1988136808a138d75e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 13:26:23.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/97283932cd174e48b57134759617c7f7 : SUCCESS in 19m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/95c910840d4743129b79f1d07b1bdeaa : SUCCESS in 20m 00s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"cc2baf969af927e2bf4c564cfcb54d677435256c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-09 13:31:59.000000000","message":"Patch Set 21:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1654779869903 : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"db5a865f632e5b41d779a7b113b876ad6cd56116","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-09 13:38:00.000000000","message":"Patch Set 21:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-test/77","accounts_in_message":[],"_revision_number":21},{"id":"3765a4a41601aa8435dbba17a92379e4bbcc8f4f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-09 13:39:40.000000000","message":"Patch Set 21:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/21/iscsi/ : FAILURE in 34m 45s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/21/smb/ : FAILURE in 34m 41s","accounts_in_message":[],"_revision_number":21},{"id":"a2660cf9d7416c5efba5df6c19a2f463ccdc02bd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-09 13:50:10.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d45674784ef44a3e9b77dff233fbedac : SUCCESS in 42m 54s","accounts_in_message":[],"_revision_number":21},{"id":"c0b75588e56e62e082482168c406816323bbf5e1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-09 14:37:06.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k-focal https://hpe-storage.github.io/alletra-6k-ci-logs/logs/92/825492/21/check/iscsi-driver-alletra-6k-focal/36eb70d/36eb70d.html : FAILURE in 1h 31m 44s","accounts_in_message":[],"_revision_number":21},{"id":"48fc257343112743b98d01db3f70b1d9764dd70f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-06-09 14:53:53.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/825492 : SUCCESS in 1h 31m 07s","accounts_in_message":[],"_revision_number":21},{"id":"03a20fe11f565ff9249feea7a104821f4a4a5066","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-09 15:11:34.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/92/825492/21/check/dsvm-tempest-cisco-zonemanager-job/1a7ecd1/ : SUCCESS in 1h 08m 33s","accounts_in_message":[],"_revision_number":21},{"id":"4bfbbfda99b29f0383b3957e388265bdd80d0d36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 15:43:04.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f629437bb3bc4b6f8d24cbc88ca49a16 : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c29a07788f394ed5acb6bc28a2c2d815 : SUCCESS in 11m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ef97a9301754603b2bbcca0996c657d : SUCCESS in 18m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6fdedf61b4f04eddb90201fed2bcc82d : SUCCESS in 13m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c10a9eead3a4e51a60556abb9399126 : SUCCESS in 12m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/591a7e5bc9704d86b25205e4fce52729 : SUCCESS in 59m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ead7cb30960f4750877d1d452ef69ddd : SUCCESS in 1h 11m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/553d230371cb40aaafb942e987b0605b : SUCCESS in 44m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0dc403e9f01244e3a6853be61ebda081 : SUCCESS in 8m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/40075608be0c40528e570d1bdd6aca9e : SUCCESS in 15m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0eb3a5da96154c0db5359a77c84311fb : SUCCESS in 6m 44s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cc8ac23ef818406092e92b6dc49686c7 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f7686b9e48c3437a80fd653def2179d9 : SUCCESS in 6m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/07fae158fca34ef48fda29f4a5157ee2 : SUCCESS in 7m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b6657cb123624f71beb3dae44c9e09be : SUCCESS in 39m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/24a2487e98f1472cbdbc71e7d25d623b : FAILURE in 6m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b3e0605baeda44e79291a58e92f1dd49 : SUCCESS in 1h 17m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/28fca3110fe94ddaa019561606053fb8 : FAILURE in 1h 21m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/598c40f7606546238ba0366367e7c253 : SUCCESS in 1h 24m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a3576ef3d5754b7390e284725a6d1727 : SUCCESS in 1h 02m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d023e84a7d3f4edd901568351820f39b : SUCCESS in 1h 06m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e85e8ce19e524e2c89b488139cb0db35 : SUCCESS in 38m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/27338b72f5c6401a8ab85c813c255bc6 : SUCCESS in 1h 29m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c09162d1543941c8b77782eb59582c6d : SUCCESS in 2h 06m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64e9af50e25c41e2a965f4265c3a577c : SUCCESS in 49m 05s","accounts_in_message":[],"_revision_number":21},{"id":"80b3bf3cfcadc28bb11678f596b260418ae70e51","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-06-09 15:43:12.000000000","message":"Patch Set 21:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":21},{"id":"9520f5047e3ce32e430d08e9b8befdda440fd44e","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-06-09 15:43:32.000000000","message":"Patch Set 21:\n\nBuild failed.","accounts_in_message":[],"_revision_number":21},{"id":"bc471173c2ca9e2c2c19feb0bcd82668eeabe5f4","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-06-09 15:43:59.000000000","message":"Patch Set 21:\n\nBuild failed.","accounts_in_message":[],"_revision_number":21},{"id":"c1efac911912f23cb3e09b869325d4e0ada5f9a3","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-06-09 15:44:16.000000000","message":"Patch Set 21:\n\nBuild failed.","accounts_in_message":[],"_revision_number":21},{"id":"c8bbdaf3324e4783c83477e71f61ae1d76d69aca","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-06-09 15:44:26.000000000","message":"Patch Set 21:\n\nBuild failed.","accounts_in_message":[],"_revision_number":21},{"id":"f32fe787265ba5c68ae869e2dfbb5f2b80ae1fa1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-09 15:46:13.000000000","message":"Patch Set 21:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/825492/21/check/DellEMC_Unity_FC/4d89840/DellEMC_Unity_FC/953 : SUCCESS in 1h 08m 31s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/21/check/DellEMC_Unity_iSCSI/560c8bd/DellEMC_Unity_iSCSI/892 : SUCCESS in 1h 43m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":21},{"id":"1dd52e060c487c153986a10604f7b2ebe05e81be","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-09 16:09:39.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/21/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4a56715/ : TIMED_OUT in 3h 03m 23s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/825492/21/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/248c8a4/ : FAILURE in 1h 07m 42s","accounts_in_message":[],"_revision_number":21},{"id":"996bebaeb20a7422bcfc371a28232319a615cc00","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-09 16:19:26.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/825492/21/check/iscsi-driver-nimble/545d13f/545d13f.html : FAILURE in 1h 35m 05s","accounts_in_message":[],"_revision_number":21},{"id":"5bd120b65a0e0cbd93cf745b22d4f0655bb85e28","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-09 16:43:09.000000000","message":"Patch Set 21:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-21/ns5_nfs : SUCCESS in 3h 34m 41s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-21/ns5_iscsi : SUCCESS in 3h 38m 13s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":21},{"id":"99e3bedc5525dabcbf396a2051802758e2e3eebb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-09 16:47:49.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/21/upstream-check/cinder-tempest-plugin-ontap-fcp/901c67a/ : SUCCESS in 1h 02m 51s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/21/upstream-check/cinder-tempest-plugin-ontap-iscsi/4d34ee4/ : FAILURE in 51m 01s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/21/upstream-check/cinder-tempest-plugin-ontap-nfs/7b86203/ : FAILURE in 43m 08s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":21},{"id":"6d62873ae453bd31eed3408875a0e0e7070439cd","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-06-09 16:50:11.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/21/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/9e0967e/ : FAILURE in 1h 05m 14s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":21},{"id":"7196121e0e4c7fe8a6e72fec7cf1d99c60999660","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-06-09 17:36:21.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/21/check/DellEMC_PowerStore_iSCSI/1c92f56/DellEMC_PowerStore_iSCSI/567 : FAILURE in 1h 30m 27s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/92/825492/21/check/DellEMC_PowerStore_FC/c91dd29/DellEMC_PowerStore_FC/606 : FAILURE in 1h 52m 53s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":21},{"id":"9d92397cc82cacf3713ebecca97ccd96d54b031b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-06-09 17:37:38.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/92/825492/21/check/DellEMC_SC_FC/7302228/DellEMC_SC_FC/673 : SUCCESS in 46m 11s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/21/check/DellEMC_SC_iSCSI/bd9b063/DellEMC_SC_iSCSI/694 : FAILURE in 1h 54m 13s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":21},{"id":"3123f99c1f03fd6ccc9ccd326d58010a5e29f331","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-06-09 17:43:36.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/92/825492/21/check/DellEMC_PowerFlex/859204a/DellEMC_PowerFlex/846 : ABORTED in 2h 00m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":21},{"id":"b7ee86bfe040e198175efec3b47763d24854b8bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-06-09 20:00:00.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"0b0573910f8678e4154e072cf4ded7fe773952f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 20:20:53.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/708839daedf44567bb9775c37ff4cb90 : SUCCESS in 17m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f017d4d19f354364baff717c40c035cc : SUCCESS in 19m 02s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"dc816cff441ce3123dde45aa94791c759b07798a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-09 20:28:01.000000000","message":"Patch Set 22:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1654804810576 : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"a379c89be6290d5c1a8f98e405133f20a4639e99","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-09 20:30:20.000000000","message":"Patch Set 22:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-test/78","accounts_in_message":[],"_revision_number":22},{"id":"e4c543c6b35bbdff3c2e5cabe368e241d8a13fcf","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-09 20:42:27.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8ea61d049caf4ba19f068b0333ad9952 : SUCCESS in 41m 22s","accounts_in_message":[],"_revision_number":22},{"id":"561d0f95072d6325d7d4551639bcef5bce3cc9be","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-09 21:15:38.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/92/825492/22/check/dsvm-tempest-cisco-zonemanager-job/24dab50/ : SUCCESS in 1h 07m 43s","accounts_in_message":[],"_revision_number":22},{"id":"f83239a938a4ce9e311dd4d24ad72a95d510e5b2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-06-09 21:22:26.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/825492 : SUCCESS in 1h 20m 46s","accounts_in_message":[],"_revision_number":22},{"id":"d85509dc08a1c3a7ee40d2ae8b214c45b57c1c43","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-09 21:30:23.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k-focal https://hpe-storage.github.io/alletra-6k-ci-logs/logs/92/825492/22/check/iscsi-driver-alletra-6k-focal/a24031d/a24031d.html : FAILURE in 1h 29m 20s","accounts_in_message":[],"_revision_number":22},{"id":"ce6da50ce64091b6957aeabf140231ac37d605f0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-09 21:44:47.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/825492/22/check/DellEMC_Unity_FC/112296d/DellEMC_Unity_FC/955 : SUCCESS in 1h 14m 23s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/22/check/DellEMC_Unity_iSCSI/058a84e/DellEMC_Unity_iSCSI/894 : SUCCESS in 1h 44m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":22},{"id":"c616f2b54514d00bc9f12c17cf017d4842a299d2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-09 21:47:01.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/825492/22/check/iscsi-driver-nimble/5eb9040/5eb9040.html : FAILURE in 1h 45m 49s","accounts_in_message":[],"_revision_number":22},{"id":"0a3eeedb11dc55253950e786b971ec9f6b437110","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 22:19:32.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af510d3eba3c4c52a3d4e602201b259c : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4699548ba5546bdaff7b1508bebaf10 : SUCCESS in 10m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d980c75a93b4977ba8bd5fe9872a0cb : SUCCESS in 10m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19d2df60caad4be6bbd5f03eaafc8a44 : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a71dfa3e564943eda0711e87cd9322ef : SUCCESS in 11m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/50d3d6065b6d4b409d5167270dbb67fd : SUCCESS in 50m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6c833e0bfeaa4a2fadc4dfe7712e1f60 : SUCCESS in 1h 09m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4469d4a3eb584e87ae6fc55b2acf553d : SUCCESS in 45m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f30f9c505330420ea5e6323b272ff621 : SUCCESS in 8m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1153bd3ecfb9428f8cdf6aca22706c81 : SUCCESS in 12m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e22402a722544f81bf3f199875475623 : SUCCESS in 5m 29s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2a88334fef454ec5ac36ab104772702a : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/86d7b09134db45a68c1bf59e7693c879 : SUCCESS in 6m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/82848a91c5334ca2b9b3ac8c5d8f7a8c : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/91b53efd6555478f98cb2b5ec10827d8 : SUCCESS in 34m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bee5bae7c444470db50f0109f525abc2 : FAILURE in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/045a6ee4c4a5493b96c03bb0cb420039 : SUCCESS in 1h 05m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/02b7a51bfe384404a5c8e7ef38137eee : FAILURE in 1h 29m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/98f2b7ec35e0440293a928cdfa7df211 : SUCCESS in 1h 21m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7cdfbd8cdc724928a539b7b990ed7ff4 : SUCCESS in 1h 00m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f96d1f7c546b4f18b63bf3966f5ead39 : SUCCESS in 1h 08m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/52a695f1686549a89e217e965cf6e367 : SUCCESS in 43m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/be04af027e184295b4de866ae4e85acd : SUCCESS in 1h 11m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ede2d738cf364d7a9f3abc4427120f1e : SUCCESS in 2h 11m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd7089b751ae41f29eb882e83fe6e61f : SUCCESS in 45m 23s","accounts_in_message":[],"_revision_number":22},{"id":"e9ba940b17e358ffde778931fce14e5d89765199","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-06-09 22:19:42.000000000","message":"Patch Set 22:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":22},{"id":"e00aedad72469c079a17052e144f55ec95e77ecc","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-06-09 22:20:19.000000000","message":"Patch Set 22:\n\nBuild failed.","accounts_in_message":[],"_revision_number":22},{"id":"4ac5da1647744250dce36f506f9cddb8617b0a63","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-06-09 22:20:21.000000000","message":"Patch Set 22:\n\nBuild failed.","accounts_in_message":[],"_revision_number":22},{"id":"f2b01e012c401395ae10f8945c89687b378b853a","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-06-09 22:20:46.000000000","message":"Patch Set 22:\n\nBuild failed.","accounts_in_message":[],"_revision_number":22},{"id":"dbff1bcdc1fbfd18806ad3bdf9624ca714e592de","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-06-09 22:21:12.000000000","message":"Patch Set 22:\n\nBuild failed.","accounts_in_message":[],"_revision_number":22},{"id":"5c73e4dc44de2487f017d964edf92558919d9722","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-09 23:04:59.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/22/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/720b53c/ : TIMED_OUT in 3h 03m 18s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/83b4d326817242d0a9119b2fd778e997 : DISK_FULL in 2h 54m 44s","accounts_in_message":[],"_revision_number":22},{"id":"036408ee472438c3b04146e80855cb86f448c906","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-09 23:22:26.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/22/upstream-check/cinder-tempest-plugin-ontap-fcp/1213080/ : SUCCESS in 1h 02m 19s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/22/upstream-check/cinder-tempest-plugin-ontap-iscsi/48f25d5/ : FAILURE in 46m 22s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/22/upstream-check/cinder-tempest-plugin-ontap-nfs/ae5664f/ : FAILURE in 42m 56s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":22},{"id":"7eea23896211a656dd1d725e585b05785accd5c0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-09 23:30:21.000000000","message":"Patch Set 22:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-22/ns5_nfs : SUCCESS in 3h 18m 23s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-22/ns5_iscsi : SUCCESS in 3h 29m 52s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":22},{"id":"44b82d137f08d285dbe92082f7b8c008236f74e1","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-06-09 23:46:45.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/22/check/DellEMC_PowerStore_iSCSI/9dbeeb3/DellEMC_PowerStore_iSCSI/568 : SUCCESS in 1h 20m 52s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/92/825492/22/check/DellEMC_PowerStore_FC/3c1ad6d/DellEMC_PowerStore_FC/607 : SUCCESS in 1h 26m 49s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":22},{"id":"7b07d9328bdd845d97c4ccb2982da78b0c8e4dd5","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-06-09 23:55:11.000000000","message":"Patch Set 22:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/92/825492/22/check/DellEMC_SC_FC/2ae4c34/DellEMC_SC_FC/674 : SUCCESS in 55m 17s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/22/check/DellEMC_SC_iSCSI/796c70b/DellEMC_SC_iSCSI/695 : FAILURE in 1h 35m 18s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":22},{"id":"c31f4f15b380304c13c5b86d697b0c24dd8be199","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-06-10 00:19:56.000000000","message":"Patch Set 22:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/92/825492/22/check/DellEMC_PowerFlex/9e7a472/DellEMC_PowerFlex/847 : ABORTED in 2h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":22},{"id":"fe89382bd1e3e1354ac24e85b24ab531702e98d0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-10 00:58:40.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/825492/22/check/DellEMC_Unity_FC/4ff16a2/DellEMC_Unity_FC/957 : SUCCESS in 1h 09m 29s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/22/check/DellEMC_Unity_iSCSI/9244b35/DellEMC_Unity_iSCSI/896 : SUCCESS in 1h 40m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":22},{"id":"7aba23206266ae10b2365d873220a92d852c8526","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-12 14:03:35.000000000","message":"Patch Set 22:\n\nUNSTABLE build log link http://tempest-openstack.spb.yadro.com/gerrit-test/120","accounts_in_message":[],"_revision_number":22},{"id":"0bb600436f893308d4d14ccf2d15cb79cfe261ca","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-13 05:28:11.000000000","message":"Patch Set 19:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":19},{"id":"90a6bf52261b9d02fedb93b6ec006413425ec305","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-13 05:29:55.000000000","message":"Patch Set 20:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":20},{"id":"bc89df81ca413fc9f4199598c507bd974303d4e4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-13 05:31:05.000000000","message":"Patch Set 22:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":22},{"id":"8c10fa6a150f76a884558dff0f04ff315ab14be3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-13 05:31:25.000000000","message":"Patch Set 21:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":21},{"id":"b2cc7e5da4919608e7608ed3d0334716338d6f40","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-15 18:25:59.000000000","message":"Patch Set 22:\n\nUNSTABLE build log link http://tempest-openstack.spb.yadro.com/gerrit-test/127","accounts_in_message":[],"_revision_number":22},{"id":"5b0dd48262b53b20debfd5172ce68c650ee2df03","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-17 12:09:23.000000000","message":"Patch Set 22:\n\nUNSTABLE build log link http://tempest-openstack.spb.yadro.com/gerrit-test/128","accounts_in_message":[],"_revision_number":22},{"id":"521f6c383da2cfdf8a29161e6cff3bbd652846ca","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2022-06-17 12:20:16.000000000","message":"Patch Set 22: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"bfd4a043a281fd50e6121e080c950161a8ed0b1c","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-06-17 13:10:22.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"400d7e1bf908d533f4b6cc3bfc6bb2075ff1d79c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-06-17 14:22:22.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"a8cf4f3aabb4d0b7a5f68d8027d90fdbb1905c87","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-06-17 14:24:02.000000000","message":"Patch Set 23:\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"c075aaaaf99ab9f60cca94984daa15f829e50987","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 14:54:58.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f60472e5eca943509977ad65ab5a5779 : SUCCESS in 22m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a0350c5ec36347f69dd01b323a0ee938 : FAILURE in 25m 22s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"7c1a70ed421578a97f7cd818fc75423405a804f1","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2022-06-17 15:02:10.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"d34b457d3f14381568fa3854be910ebc529ae2c5","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-06-17 15:14:04.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"155a2e5e44683abd0ae055f1b888c2b89cb11638","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-17 15:54:11.000000000","message":"Patch Set 23:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/23/iscsi/ : FAILURE in 1h 31m 09s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/23/smb/ : SUCCESS in 1h 04m 46s","accounts_in_message":[],"_revision_number":23},{"id":"fa30787bbf1c3ca06d79d7c73cbecc6e7dc593dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 16:49:36.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9aa9cef612e4e50908a7c8517bec908 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a18f5c530f3c472eadad58a5718ad90b : SUCCESS in 11m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f61aee09a7d84a859e196cda1fe4b0c7 : SUCCESS in 11m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf4d95b8e0a6467082de036fc8aa5c46 : SUCCESS in 11m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d80f6ec087d4166bd7c0ac0fecec42d : SUCCESS in 15m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/246d5965196f43beaece425a82dcf741 : SUCCESS in 54m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a3a45631c93147eda862559a1fe360ed : SUCCESS in 1h 19m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f80b0300a7a4ba3be75a12d35bc8628 : SUCCESS in 44m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/590d0d8c587e439d8599c1e91b7d92f0 : SUCCESS in 8m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4f56f4acec1d499eaa20fe8ab3cf8b00 : SUCCESS in 16m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cef40d310cdb4f70bb14dfb997777880 : SUCCESS in 5m 56s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c05745dac7a74943b5a602464a6c9ad6 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/98d0f3e05a3c4f0fbc5ed1c10ad79d0b : SUCCESS in 8m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7a36ed7c77d74c35834c9638a21eb468 : SUCCESS in 6m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/894d2dd6aae844b7b1dcdbafce0f8425 : SUCCESS in 43m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8403d88101a94f2383336e7d430a9efc : FAILURE in 6m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/400b3035637e479c8ab1530f1f42c39a : SUCCESS in 1h 14m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/96733704be4c452d9c4b75d72c600c58 : FAILURE in 1h 33m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e35402e9fd2242d098ac07858fef01cf : SUCCESS in 1h 35m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/08177f614aaf4ff3a7bbe6e9aae7cadd : SUCCESS in 55m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9f3c48cf8aeb4c5289b242bb9a0c2ed8 : SUCCESS in 1h 11m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/24e3711770d649aa9262900a4fb8ead5 : SUCCESS in 37m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d929e22c032d4267812cfb7035b02210 : FAILURE in 1h 04m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/80ef72d94f374740b42a654b8cb3a993 : SUCCESS in 2h 18m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1924a90ea1b4134b5954774ee008020 : SUCCESS in 42m 23s","accounts_in_message":[],"_revision_number":23},{"id":"9c60f8ae6957a8a2413b3182961f928c55155778","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-06-17 16:49:45.000000000","message":"Patch Set 23:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":23},{"id":"a4ce5b24d86dee9534539d8f57c466a6a2351d33","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-17 17:58:22.000000000","message":"Patch Set 23:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-23/ns5_nfs : SUCCESS in 3h 24m 58s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-23/ns5_iscsi : SUCCESS in 3h 35m 28s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":23},{"id":"29297b9a0b62b16e9e6812ffa3214e7d24fe4729","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-17 17:59:05.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k-focal https://hpe-storage.github.io/alletra-6k-ci-logs/logs/92/825492/23/check/iscsi-driver-alletra-6k-focal/c4edc42/c4edc42.html : FAILURE in 1h 31m 15s","accounts_in_message":[],"_revision_number":23},{"id":"d8eaedc1d7b2cb3c66d2d7586b653ff38b2eed95","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-17 18:02:02.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/23/upstream-check/cinder-tempest-plugin-ontap-fcp/68c4b4f/ : SUCCESS in 1h 00m 47s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/23/upstream-check/cinder-tempest-plugin-ontap-iscsi/359e00f/ : FAILURE in 1h 10m 42s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/23/upstream-check/cinder-tempest-plugin-ontap-nfs/39d12ea/ : SUCCESS in 1h 06m 56s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":23},{"id":"8abc768e92424d30c6bfb1335ca7f49a237237c5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-06-17 18:21:56.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/92/825492/23/check/DellEMC_PowerFlex/71a7800/DellEMC_PowerFlex/867 : FAILURE in 1h 31m 54s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":23},{"id":"98ba6688203b6eacf0ced1b5760e6d23d3834e93","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-17 18:29:48.000000000","message":"Patch Set 23:\n\nUNSTABLE build log link http://tempest-openstack.spb.yadro.com/gerrit-test/129","accounts_in_message":[],"_revision_number":23},{"id":"68efaedaae80d36e0c16cfd22bd507e76a879f51","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-17 19:43:09.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/66f251b113f847849281ab35bc2658fd : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":23},{"id":"aefb4aa962a2853d1ec9eb2a7234f0ad3335407c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-17 20:04:52.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/23/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/3d81a6e/ : TIMED_OUT in 3h 02m 55s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/825492/23/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/4630624/ : FAILURE in 42m 22s","accounts_in_message":[],"_revision_number":23},{"id":"0b13f858003f3ef8817a7c600c0a03a096743303","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-17 21:14:20.000000000","message":"Patch Set 23:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/5fcb0f576e0b4ecd88899b40635c50b3 : POST_FAILURE in 1h 42m 06s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/bd24fe911dca48dab65f146e2f626b50 : POST_FAILURE in 2h 48m 05s","accounts_in_message":[],"_revision_number":23},{"id":"bc37b3375b9e35e2f9cb164fceb2b3e0d49910ab","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-17 23:54:42.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/825492/23/check/iscsi-driver-nimble/a813194/a813194.html : FAILURE in 1h 00m 13s","accounts_in_message":[],"_revision_number":23},{"id":"f6a74fcb49a1c3b38c81bdf12c6b3d7f20729abc","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-06-18 10:19:35.000000000","message":"Patch Set 23:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/825492-23/HPMSA-iSCSI/ : FAILURE in 54:33.85\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/825492-23/HPMSA-FC/ : FAILURE in 51:09.26","accounts_in_message":[],"_revision_number":23},{"id":"fe1f55e5b8cc0582e068fadffaafb989e774d534","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-06-18 14:05:54.000000000","message":"Patch Set 23:\n\nBuild failed.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/825492-23/PVME-iSCSI/ : FAILURE in 1:03:56\n- PVME-FC https://os-logs.seagate.com/dell-ci/825492-23/PVME-FC/ : FAILURE in 1:02:21","accounts_in_message":[],"_revision_number":23},{"id":"99e7a77ed1dc361838bb6b3779623081b3a7a0a4","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-06-18 14:39:47.000000000","message":"Patch Set 23:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/825492-23/Lenovo-iSCSI/ : FAILURE in 1:06:02\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/825492-23/Lenovo-FC/ : FAILURE in 1:05:07","accounts_in_message":[],"_revision_number":23},{"id":"b6904e0da169b2a760bc02e6f035465bfcd5213d","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-06-18 15:53:56.000000000","message":"Patch Set 23:\n\nBuild failed.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/825492-23/Seagate-iSCSI/ : FAILURE in 1:07:26\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/825492-23/Seagate-FC/ : FAILURE in 1:06:38","accounts_in_message":[],"_revision_number":23},{"id":"ff10aadb24c49211a58b514f8ccca0861b1a3196","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-18 16:14:50.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/92/825492/23/check/dsvm-tempest-cisco-zonemanager-job/e44ca3e/ : SUCCESS in 1h 11m 13s","accounts_in_message":[],"_revision_number":23},{"id":"895346956f2c0b86199ecc13974b05883e96e0c6","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-06-19 05:04:28.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/92/825492/23/check/DellEMC_SC_FC/3991d86/DellEMC_SC_FC/712 : ABORTED in 3h 00m 01s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/23/check/DellEMC_SC_iSCSI/c367ef7/DellEMC_SC_iSCSI/733 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":23},{"id":"0cae2d84d80eee47098de3f5855aa25ef6fa368a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-06-19 07:59:13.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/92/825492/23/check/DellEMC_PowerMAX_FC/a531eae/DellEMC_PowerMAX_FC/586 : SUCCESS in 3h 35m 15s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/23/check/DellEMC_PowerMAX_iSCSI/84436ea/DellEMC_PowerMAX_iSCSI/573 : SUCCESS in 3h 32m 49s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":23},{"id":"63f4193041488510e10cb0940969302125e9abad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-19 11:19:47.000000000","message":"Patch Set 23:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-test/137","accounts_in_message":[],"_revision_number":23},{"id":"957ce547c887b684b61e5fba42706f2c2d52f3f6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-20 05:26:15.000000000","message":"Patch Set 23:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":23},{"id":"f2b1e39d70a3dde1ed411c0028a6242902df2a14","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2022-06-21 08:43:46.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"c4ba59ee55c7e79cd5aedff3e8bf06ab1f25c285","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-06-21 10:08:34.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/92/825492/23/check/DellEMC_PowerFlex/db47e1e/DellEMC_PowerFlex/873 : SUCCESS in 1h 24m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":23},{"id":"457ff66209c0d00dc86a1e3cd7e6db2818cd3d7d","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-06-21 10:33:03.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"ed00c768099184c0b04a716953453a950c75e914","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-06-21 10:44:18.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"a1df7a83d09d1b4f9eb2eaf9a34390b691152655","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-06-21 15:29:54.000000000","message":"Patch Set 23:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-test/138","accounts_in_message":[],"_revision_number":23},{"id":"bb25b21007f8896352d75b740755cdcdc3b83b3b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-22 08:31:04.000000000","message":"Patch Set 23:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/825492/825492/23/check-not-vote/ext-cinder-zuul/c6b1b61 : FAILURE in 16m 58s","accounts_in_message":[],"_revision_number":23},{"id":"85a97cfb3014da6fbafc3480d31f6870da94b6c0","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2022-06-27 03:00:19.000000000","message":"Patch Set 18:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/92/825492/18/fc : SUCCESS in 1h 13m 45s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/92/825492/18/iscsi : SUCCESS in 1h 30m 43s","accounts_in_message":[],"_revision_number":18},{"id":"da09f8f5a7be4e47c765a46d5f968edfe29b5646","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-06 11:27:28.000000000","message":"Patch Set 23:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/140","accounts_in_message":[],"_revision_number":23},{"id":"70337d1c193e906a7594a91a98055194ba282b06","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-06 14:28:17.000000000","message":"Patch Set 23:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-test/141","accounts_in_message":[],"_revision_number":23},{"id":"661514aab06dfc070992101c8c7e6aa29f03a590","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-11 06:43:47.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"60f0ff615f1bbca486b8ea6b714a7f1d59b7a67c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-11 06:43:57.000000000","message":"Patch Set 24:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16154/","accounts_in_message":[],"_revision_number":24},{"id":"5b1311e349005c64abe4735f48d3f2e143dff472","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-11 06:47:07.000000000","message":"Patch Set 24:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1657521837871 : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"13ca1ab6ecf53e26ee085db2432b28bada1cdb77","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-11 06:47:10.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"9554a5bd34066439a236ab17e3bf26c8baefaa65","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-11 06:47:41.000000000","message":"Patch Set 24:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16154/ : FAILURE","accounts_in_message":[],"_revision_number":24},{"id":"4661d18f71a6ba9483400bdcef42979489686b85","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-11 06:52:41.000000000","message":"Patch Set 24:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/825492/825492/24/check-not-vote/ext-cinder-zuul/c62375e : FAILURE in 4m 25s","accounts_in_message":[],"_revision_number":24},{"id":"ced1c2ddd2fba90213d8aab12961ce2172d2c2db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 07:03:33.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b249a749f7424ea1acc8332c75c07c4e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/161ffea56c974196b13768ac8d70aa82 : SUCCESS in 14m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b9f93e24461843e9bc895019e4bea6cc : SUCCESS in 14m 41s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"749f8266d968f6a5b21bb4a0900c31df93de7eae","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-11 07:45:56.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/92/825492/24/check/dsvm-tempest-cisco-zonemanager-job/c258c45/ : FAILURE in 51m 43s","accounts_in_message":[],"_revision_number":24},{"id":"3fd3b164e835892a120850f6c9f5944370f1114e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-11 07:47:36.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/825492/24/check/iscsi-driver-nimble/e9fae99/e9fae99.html : FAILURE in 1h 02m 28s","accounts_in_message":[],"_revision_number":24},{"id":"da66048b69af44422b7279045a3c014f0e6130f4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-11 07:50:24.000000000","message":"Patch Set 24:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/148","accounts_in_message":[],"_revision_number":24},{"id":"89677583e8d4cd806adecef61f473161d3185877","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-11 07:55:11.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/24/iscsi/ : SUCCESS in 1h 10m 50s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/24/smb/ : SUCCESS in 1h 07m 18s","accounts_in_message":[],"_revision_number":24},{"id":"8a9dbd7fa25ec899c8d5e2a62ef316fbf6b360d3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-11 07:56:14.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/825492 : FAILURE in 1h 10m 48s","accounts_in_message":[],"_revision_number":24},{"id":"78463262a45ce869775f512f29e0b13d91e17546","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-11 08:18:20.000000000","message":"Patch Set 24:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/24/check/tempest-dsvm-full-focal-py3/e8a51ec/ : SUCCESS in 1h 20m 53s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/24/check/devstack-plugin-ceph-dsvm-focal-py3/7da9673/ : SUCCESS in 1h 31m 17s","accounts_in_message":[],"_revision_number":24},{"id":"83fb2e528911b365715258db4b24cf4ba55bc79f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-11 08:28:16.000000000","message":"Patch Set 24:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/92/825492/24/check/fujitsu-eternus-dx-iscsi/7c89636 : SUCCESS in 1h 43m 49s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/92/825492/24/check/fujitsu-eternus-dx-fc/36e3a51 : FAILURE in 12s","accounts_in_message":[],"_revision_number":24},{"id":"8798430de922d9a688643895cca563e12ba67ff8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-11 08:33:54.000000000","message":"Patch Set 24:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/825492/24/check/DellEMC_Unity_FC/1d8b6ff/DellEMC_Unity_FC/1120 : SUCCESS in 1h 10m 09s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/24/check/DellEMC_Unity_iSCSI/2809278/DellEMC_Unity_iSCSI/1052 : SUCCESS in 1h 49m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":24},{"id":"31706a117f39648e11e07674ea973b6ea05e5ecc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 09:15:59.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6dfd408ade414340b19b4013195681b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bb1a856845249978a16fb8978b53c32 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92260ad9a9b844019ac8d94a76bb7144 : SUCCESS in 11m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43ae83d74f3f4b47b9fa7e1cf08c0f4d : SUCCESS in 11m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46a7a73baa1f43898b99fcd36e40191b : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b4b8a9b948842c0a8de879264bc5cf8 : SUCCESS in 12m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/9003c6f0012547feb823dff0ffa1dc4b : SUCCESS in 1h 04m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e78e4d105727463e8c3f6ee0cebc4822 : FAILURE in 1h 28m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3cc498f06eb94aeba0df63295b6640db : SUCCESS in 55m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b778bf6941414d918193026d7fbe3a8c : SUCCESS in 8m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/66c2524fdeaa4710b2003b456f988bd9 : SUCCESS in 15m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/98c6b23ce7c34c939829d69cfaf280a9 : SUCCESS in 5m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/24c9022fc61c43a2a4ea07a34c7942f6 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6caea46e62614f1383b39ae040e8cd07 : SUCCESS in 5m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dbd6f18619a543b08474e47067b18ce5 : SUCCESS in 6m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/58ba7084100845cfbfe404fec6e3cd76 : SUCCESS in 39m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/83a2b72110df4a309198fe3cb2ccb343 : FAILURE in 5m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e9a9f043a58148bd8b1df35b573bee1b : SUCCESS in 1h 19m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aa7b00a171604ecd818026e6d2090739 : FAILURE in 1h 11m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/988098f2a0af45d9ad21e3e5070802f3 : SUCCESS in 1h 35m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e888f453f45a4718baaf56c82a67011a : SUCCESS in 50m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d6885051405a4afdac4631d22f9fb609 : SUCCESS in 1h 48m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4548758b42c74eb88ea0078604e2cb96 : SUCCESS in 35m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d60609f79eb24a2492731d748bbdf784 : SUCCESS in 1h 16m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d695cc8e8eff43bfa1e6619196b97f45 : SUCCESS in 2h 25m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a57004bb8be54d27b7a3bf43ff22ea65 : SUCCESS in 36m 52s","accounts_in_message":[],"_revision_number":24},{"id":"e739c006e2f5e18b3a755442ff581b95030fab23","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-11 09:17:46.000000000","message":"Patch Set 24:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/149","accounts_in_message":[],"_revision_number":24},{"id":"3a7378c3f47e6b75a26424ef646aff30f128d4b0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-11 10:02:37.000000000","message":"Patch Set 24:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/150","accounts_in_message":[],"_revision_number":24},{"id":"ce1714715ffb71707eeb51dcee3105f6d9445f98","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-11 17:05:53.000000000","message":"Patch Set 24:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-test/151/ : ABORTED","accounts_in_message":[],"_revision_number":24},{"id":"5bc9389fd551842e456d5c9022e12cce23a70c6b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-07-11 18:32:57.000000000","message":"Patch Set 24: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":24},{"id":"69ca33a491248f517765af63c8a8ec1b3a1e9c76","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-11 20:21:00.000000000","message":"Patch Set 24:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-test/152","accounts_in_message":[],"_revision_number":24},{"id":"aab725dc9fe7f738bcce5e5017e5ee9f7d7292bd","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-12 05:46:45.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"d09000fa04ac0ee62addcddb9f1e4ff346338121","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-12 06:30:41.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"5f8aabfea816934baf0ec3875791ac41fd0c8785","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-12 06:31:06.000000000","message":"Patch Set 25:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":25},{"id":"fbb993517d55ce730857dc0ae27bb9ce58c82cad","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-12 06:31:44.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"3540a10cf22dc4d8020661ed39a30100deb14753","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-12 06:33:59.000000000","message":"Patch Set 25:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1657607450183 : FAILURE","accounts_in_message":[],"_revision_number":25},{"id":"16881f5076d0aced4701c06945ddee5fd8383f1e","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-12 06:40:12.000000000","message":"Patch Set 25:\n\n(10 comments)","accounts_in_message":[],"_revision_number":25},{"id":"5e8f6d095e9da5f0d32ac95559ec3f9eeb39eaf2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 06:51:01.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4ff8fd94ec64f66a1778af42d476d62\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7ffd28f4cee4a0f9b3f574da7e4c329 : SUCCESS in 15m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/27fc4713ab1440d69e405aca652cc601 : SUCCESS in 15m 50s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"c8065607555d0d10a2965e9a735e0926f662f8e0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-07-12 07:09:57.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4295138dba9f40059b3a69ff0516fc63 : SUCCESS in 37m 59s","accounts_in_message":[],"_revision_number":25},{"id":"34ad89dc1ce393b550a4403e859cb7480f49a3d0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-12 07:26:11.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/92/825492/25/check/dsvm-tempest-cisco-zonemanager-job/725d9b0/ : FAILURE in 50m 48s","accounts_in_message":[],"_revision_number":25},{"id":"2e727539789e4bec797b03993655bdeb6427a8a6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-12 07:38:02.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/25/iscsi/ : SUCCESS in 57m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/25/smb/ : SUCCESS in 1h 06m 47s","accounts_in_message":[],"_revision_number":25},{"id":"77dab9d684f9c581e15e03f64f10435f341cd4e5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-12 07:52:09.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/825492 : FAILURE in 1h 19m 51s","accounts_in_message":[],"_revision_number":25},{"id":"1c080ad3fd2f923d0a3f7ef5f5c7036cbea773f5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-12 08:14:32.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/825492/25/check/iscsi-driver-nimble/7cf0a22/7cf0a22.html : FAILURE in 1h 42m 25s","accounts_in_message":[],"_revision_number":25},{"id":"f163b4288cdcb2d0955b387b5de61c279109cdba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 08:32:51.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fdf27992a11418d95a78cb18f3d6af5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f73324a51f34cfca48f60a9ae0088db : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08c1cc4285b8411182cdccb313aae0da : SUCCESS in 10m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7c7c40132084fb093697ed834d13b32 : SUCCESS in 10m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22da0317438a4779957840fb9125417a : SUCCESS in 16m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a99b21cee065423a807389089974dcc9 : SUCCESS in 10m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/d6cc1640a5c5478999312685f09742db : SUCCESS in 1h 03m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a451111d1b194e17ad10843a4b45900f : SUCCESS in 1h 13m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3dec36daf26e47f5b87db14a9ffbc4ba : SUCCESS in 1h 03m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f6cd798622e467c80b03d12efaeeafb : SUCCESS in 7m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4ed0c6fb1973421191d9ab356bd06c6f : SUCCESS in 14m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/665ce4f1c7bd4a3199a6d8413593e29f : SUCCESS in 8m 02s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3643891270d14d15bdcb6ec8da184856 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/53e974c946ea4f13a39ac1e9375b941a : SUCCESS in 6m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3d30804ceb9e42a69583eec1aad8ebb3 : SUCCESS in 7m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/06cb6ada69874c60a2eeb87de22abf0b : SUCCESS in 52m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f6fd4a552abe4dd7aa366837574e1ce4 : FAILURE in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/569dbd82f759458da9fe59c7ebbc5500 : SUCCESS in 1h 22m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/236f6b6a7c4a452fb0e1857321e03590 : FAILURE in 1h 37m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4b487e5984e5409a9a38a24c2313262b : SUCCESS in 1h 40m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7d1b3e3a67064a628e4a43e0688902a9 : SUCCESS in 1h 09m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8e2dc478cb75478aa4e021a3b7401bf5 : SUCCESS in 1h 06m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6b7f80b015084aed9f7ac1c26910e64e : SUCCESS in 53m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/99d158893d16454ea797bb5cb02b714f : FAILURE in 1h 12m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c4e60d65d07d4dd88646a4d8857cc1c0 : SUCCESS in 1h 59m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46480ad6f9cc41bba0756b9a65f33b2c : SUCCESS in 56m 26s","accounts_in_message":[],"_revision_number":25},{"id":"05bb766fd1b9345e10d490f51f207fc492271580","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-12 08:35:40.000000000","message":"Patch Set 25:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/825492/25/check/DellEMC_Unity_FC/550b7b5/DellEMC_Unity_FC/1131 : SUCCESS in 1h 11m 04s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/25/check/DellEMC_Unity_iSCSI/91a7d23/DellEMC_Unity_iSCSI/1062 : SUCCESS in 1h 49m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":25},{"id":"2f559783fb5277c9a19fd92646cdbf903481e49f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-12 08:50:57.000000000","message":"Patch Set 25:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/25/check/tempest-dsvm-full-focal-py3/84c280c/ : FAILURE in 1h 24m 46s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/25/check/devstack-plugin-ceph-dsvm-focal-py3/ddb01aa/ : FAILURE in 2h 17m 07s","accounts_in_message":[],"_revision_number":25},{"id":"a12b7add4d4f0cdfed337d1b0b08f1160e5d53d6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-12 09:16:39.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp cinder-tempest-plugin-ontap-fcp : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/25/upstream-check/cinder-tempest-plugin-ontap-iscsi/eac1ded/ : FAILURE in 42m 35s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/25/upstream-check/cinder-tempest-plugin-ontap-nfs/13313dd/ : RETRY_LIMIT in 7m 21s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":25},{"id":"a43c5ed7724e3b6aa2264720ce6935a4b6d46d48","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-12 09:40:44.000000000","message":"Patch Set 25:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-test/153","accounts_in_message":[],"_revision_number":25},{"id":"8beb4a9934cab274910e1d1dacebd25b6b25defa","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-12 10:00:59.000000000","message":"Patch Set 25:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/92/825492/25/check/DellEMC_PowerFlex/9c25c0d/DellEMC_PowerFlex/972 : SUCCESS in 1h 27m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":25},{"id":"45db743bde43e1144c1cf1f9f98b048b1d5257ea","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-12 10:21:03.000000000","message":"Patch Set 25:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/92/825492/25/check/DellEMC_PowerStore_NFS/9348d25/DellEMC_PowerStore_NFS/101 : ABORTED in 21m 42s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":25},{"id":"0c4e6e5de54090744b0d82d719387771f94f504c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-12 10:24:03.000000000","message":"Patch Set 25:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-25/ns5_nfs : FAILURE in 3h 52m 52s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-25/ns5_iscsi : SUCCESS in 3h 22m 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":25},{"id":"70d284574206921b838bf1344eceb8a5ce36369c","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2022-07-12 14:45:32.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"8d23fd8b43ab810cb9963ff402d68677c2be73c2","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-07-12 15:21:42.000000000","message":"Patch Set 25:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/92/825492/25/check/DellEMC_PowerMAX_FC/2e9a7b2/DellEMC_PowerMAX_FC/708 : ABORTED in 6m 54s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/25/check/DellEMC_PowerMAX_iSCSI/5324260/DellEMC_PowerMAX_iSCSI/697 : ABORTED in 6m 51s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":25},{"id":"6c7344a3f320c055c761a4c8d13824445a58e226","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-12 15:25:08.000000000","message":"Patch Set 25:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/92/825492/25/check/DellEMC_PowerStore_NFS/ee7fece/DellEMC_PowerStore_NFS/103 : FAILURE in 39m 15s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":25},{"id":"727481db0f8265ec02f181531ea9c900f29c5490","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-07-12 21:20:31.000000000","message":"Patch Set 25: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"b03516fc568c955b5ea15f94d1bf44c860e2dc80","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-13 02:36:33.000000000","message":"Patch Set 25:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":25},{"id":"6784cb7dbf3a50ffb2a8bdf50e4158357595e062","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-13 08:40:54.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"7c756b82eafa1c5ada187476376be0dd2ed6b405","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-07-13 12:04:22.000000000","message":"Patch Set 25: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":25},{"id":"661bf34ce23c23f6b6adbfd44a52456b8f26326a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-13 12:10:08.000000000","message":"Patch Set 25:\n\nUNSTABLE build log link http://tempest-openstack.spb.yadro.com/gerrit-test/154","accounts_in_message":[],"_revision_number":25},{"id":"5af04e0f1261c9801e03b2284e859419700885de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-13 12:20:53.000000000","message":"Patch Set 25:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/155","accounts_in_message":[],"_revision_number":25},{"id":"94541081361b0bdf2ebcba46a0493e898161f43d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-13 15:35:38.000000000","message":"Patch Set 25:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-test/156","accounts_in_message":[],"_revision_number":25},{"id":"e210f46ac23918afe7bddcece3929a52092a80bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-14 08:54:29.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"a31bce7f4a236c570fe309cdbedaa8c0692ca3dd","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-14 08:55:41.000000000","message":"Patch Set 26:\n\n(7 comments)","accounts_in_message":[],"_revision_number":26},{"id":"5adbddc6b8c68c18a671177facb91edd405ee583","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 09:16:54.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/791873ef70484858a2fe88c37a02dc81\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1bdcd99fcd7b477ead30b79fdd4035e3 : SUCCESS in 15m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/09d68044c5184134a6b563693649ceea : SUCCESS in 16m 49s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"b9582541029679e83fb6ac5c64dfe42065aab461","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-07-14 09:41:17.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d3c68cda683f4f38ab028f70d079349a : SUCCESS in 45m 34s","accounts_in_message":[],"_revision_number":26},{"id":"976f091e1b070d00c726ea7ef90274cc45fa9280","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-14 10:26:23.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/26/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/743caf6/ : SUCCESS in 1h 30m 54s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/825492/26/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7f88b6e/ : SUCCESS in 1h 27m 01s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/92/825492/26/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/f09e71d/ : FAILURE in 30m 01s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"20bea6d45bd4a9f6a6924de0f596d98fe6605981","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-14 10:28:36.000000000","message":"Patch Set 26:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/26/iscsi/ : FAILURE in 1h 33m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/26/smb/ : SUCCESS in 1h 12m 09s","accounts_in_message":[],"_revision_number":26},{"id":"444720142bed664f80836ae1913e45363cddf41d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-14 10:33:53.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/825492 : FAILURE in 1h 37m 46s","accounts_in_message":[],"_revision_number":26},{"id":"b66a1dffb13af8bd3ccac706dc782f1a9803738a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-14 10:39:12.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/825492/26/check/iscsi-driver-nimble/b04ca9f/b04ca9f.html : FAILURE in 1h 43m 21s","accounts_in_message":[],"_revision_number":26},{"id":"a149a2f494059c9fb58fceb2bd9042a0197a50e8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-14 11:01:15.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/92/825492/26/check/dsvm-tempest-cisco-zonemanager-job/c2b6fcb/ : FAILURE in 55m 04s","accounts_in_message":[],"_revision_number":26},{"id":"8f8f35f48dca20b80fd9d5e8a5f57cc65890b16d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 11:22:13.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e63a8fa80d346448ea3314921fba64c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/804bbf7b435c4b3cb43da2cf67d35354 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7189805856b94278ba6f4c09432e920e : SUCCESS in 11m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f66150fdbf54881810195b3f09daf18 : SUCCESS in 12m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f373e6c7c80d4fefae71c37b493f2824 : SUCCESS in 11m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f8df321f90d46c38377357652ff4865 : SUCCESS in 12m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/8db2dae0783744eb88ad565ff37bac27 : SUCCESS in 59m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a06f53211253439fb4d9a814e847c21c : SUCCESS in 1h 20m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96de0883fd5042c4b810727700d793b4 : SUCCESS in 52m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2134ab3562d4630bda0bc5304767c92 : SUCCESS in 8m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/81f3e3dbd5dd429cbc348dbad7635a6a : SUCCESS in 15m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/542ae6637a8b4fb8aca7ea465d1bfee4 : SUCCESS in 6m 19s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/75e089f1d1794c9795f3f0ecdf9c4717 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8124f95752684ca984a28c35ab45de37 : SUCCESS in 7m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/35a05ceab19c4fdca93b110762f31325 : SUCCESS in 7m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2656950a6cd64aa6888dafe03b12f7c8 : SUCCESS in 38m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c90eaa670e41449ab887215a1e8a9171 : FAILURE in 7m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/780892ebcf354cdf9eafba130f9b1226 : SUCCESS in 1h 27m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a3f6877234474362b9f5ddf49f087c1f : FAILURE in 1h 33m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eb21ffcc8cee44c7afc7acbd916932f9 : SUCCESS in 1h 35m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cf5763211dc6442c833c4dd9f904ba12 : SUCCESS in 1h 08m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ad4b78cd3b624a6bba39ecfd1aa4dff3 : FAILURE in 1h 34m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c390420b2b5d412aa02bc1500602a587 : SUCCESS in 39m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bc14faa6547d4fe0b834b905307ba2c6 : SUCCESS in 1h 37m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/833418658c934685816e1bf08ac60712 : SUCCESS in 2h 25m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37ac7a33aa754e93a5ca125a82ade8b4 : SUCCESS in 53m 08s","accounts_in_message":[],"_revision_number":26},{"id":"f98ec1b2df61890f48eb500399267dea63b2dc18","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-14 11:31:57.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/26/upstream-check/cinder-tempest-plugin-ontap-fcp/91b9a1b/ : RETRY_LIMIT in 7m 57s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/26/upstream-check/cinder-tempest-plugin-ontap-iscsi/2d1837a/ : RETRY_LIMIT in 7m 20s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/26/upstream-check/cinder-tempest-plugin-ontap-nfs/d2c2871/ : RETRY_LIMIT in 8m 24s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":26},{"id":"3153aa6413a1704e0c972cfbdbcc7bc2dcc47981","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-14 11:32:42.000000000","message":"Patch Set 26:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/26/check/tempest-dsvm-full-focal-py3/b98fb5d/ : SUCCESS in 2h 34m 26s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/26/check/devstack-plugin-ceph-dsvm-focal-py3/47ce932/ : SUCCESS in 1h 47m 33s","accounts_in_message":[],"_revision_number":26},{"id":"1594a2fa61d16b64727593fdabdd2b3cc98add26","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-14 12:16:20.000000000","message":"Patch Set 26:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/26/check/DellEMC_PowerStore_iSCSI/a1b3dfb/DellEMC_PowerStore_iSCSI/601 : ABORTED in 14s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/92/825492/26/check/DellEMC_PowerStore_FC/6f23b72/DellEMC_PowerStore_FC/640 : ABORTED in 26m 37s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":26},{"id":"ff305e09a645ee6362f6330aef9bb7581bbf2a73","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 12:22:32.000000000","message":"Patch Set 26:\n\nUNSTABLE build log link http://tempest-openstack.spb.yadro.com/gerrit-test/157","accounts_in_message":[],"_revision_number":26},{"id":"483acddb1874649ebaf5b465307ddef466f33628","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-14 12:24:48.000000000","message":"Patch Set 26:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-26/ns5_nfs : SUCCESS in 3h 24m 45s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-26/ns5_iscsi : SUCCESS in 3h 29m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":26},{"id":"807556edb479135c7f0ada3abb5b6a21f2c8c2a2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 12:55:42.000000000","message":"Patch Set 26:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/158","accounts_in_message":[],"_revision_number":26},{"id":"b964ffd02a4542014ef3773810ae7ec59b1feb51","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 13:18:20.000000000","message":"Patch Set 26:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/71","accounts_in_message":[],"_revision_number":26},{"id":"47d84e09a141ed15e2af81f11df8bec3f6fca141","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 13:39:22.000000000","message":"Patch Set 26:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/160","accounts_in_message":[],"_revision_number":26},{"id":"a18ee64103b41239647ff061fe1bd29fe180d1dc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-14 13:49:51.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/92/825492/26/check/DellEMC_PowerFlex/2bf2658/DellEMC_PowerFlex/989 : SUCCESS in 1h 20m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":26},{"id":"5e351ff63eb1cd3976ed89ee9b7fbed7275b59ef","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 14:01:18.000000000","message":"Patch Set 26:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/161","accounts_in_message":[],"_revision_number":26},{"id":"df97100a3193589fa57775b595941a4112c0de06","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 14:21:03.000000000","message":"Patch Set 26:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/162","accounts_in_message":[],"_revision_number":26},{"id":"e0f30f58585194660d668f034dd5544ac37b98a8","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-14 15:18:15.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"c118fd159888e8197b96f4b66a4a54270f106622","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-15 07:57:54.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"2275bd72a4cd352f8b0543792d7ca32bdeea3791","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-15 11:21:26.000000000","message":"Patch Set 26:\n\nUNSTABLE build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/97","accounts_in_message":[],"_revision_number":26},{"id":"9825ed2736b679e5159aee394fbc47fc70e9d33b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-15 15:28:54.000000000","message":"Patch Set 26:\n\nUNSTABLE build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/97","accounts_in_message":[],"_revision_number":26},{"id":"6074f8903618f6a6846aa3f677855041dd7657ad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-18 01:20:59.000000000","message":"Patch Set 26:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":26},{"id":"1c5b01abb5fcd97c85c41c24ca16cde32d5bf517","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-18 11:42:24.000000000","message":"Patch Set 26:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-trigger/97/ : ABORTED\n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-test/169/ : SUCCESS","accounts_in_message":[],"_revision_number":26},{"id":"5b2bddef84f9b2bf975374564e9b5064982dbdab","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-18 12:22:02.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"6dce532493549299adb45373e11c5c4a9ed4edcc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-18 12:47:24.000000000","message":"Patch Set 26:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-trigger/135/ : ABORTED","accounts_in_message":[],"_revision_number":26},{"id":"74ae15c20b42ffd9de35d5f732e67d3a64582136","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-18 13:10:32.000000000","message":"Patch Set 26:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/136","accounts_in_message":[],"_revision_number":26},{"id":"53ff1672ff6c80a1cf6d650e89a52482b65f6fee","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-19 12:40:30.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"7258d4dbe3725fd831704c7315a22f0d3c79ac44","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-19 13:10:44.000000000","message":"Patch Set 26:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/174","accounts_in_message":[],"_revision_number":26},{"id":"0302487dcf1ae61908acd5f65076a56cec71a7fc","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-19 13:58:25.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"8e81bf0f2a55fcdd6ab1d470b1d148be59fff043","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-19 14:28:47.000000000","message":"Patch Set 26:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/176","accounts_in_message":[],"_revision_number":26},{"id":"37e1a880046989d60ec2adc0285263f3a5371ae8","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-20 08:14:31.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"4c0fc891710995dac42f02721f764ee5800cb564","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-20 08:38:00.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"cdfdccf8c29cc043bbf4a15062ca137ecb75dfba","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-20 08:40:11.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"c4b21c18d1cf0267268bdd64964dc93aefa64d25","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-20 12:44:31.000000000","message":"Patch Set 26:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/185","accounts_in_message":[],"_revision_number":26},{"id":"daaf03a080186812c497a73cb03313c35e103356","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-20 13:17:54.000000000","message":"Patch Set 26:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/170","accounts_in_message":[],"_revision_number":26},{"id":"b5f1df6eaf855ec1bc70458b9a778aa66479fd9b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-20 17:22:14.000000000","message":"Patch Set 26:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-test/170","accounts_in_message":[],"_revision_number":26},{"id":"4b61ee1b862bef4b380c5081a6c7c5a0edeedfa4","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-21 14:08:26.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/26/20220721165609/NEC-V-FC : SUCCESS in 5h 51m 32s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/825492/26/20220721165609/NEC-V-ISCSI : SUCCESS in 6h 12m 14s","accounts_in_message":[],"_revision_number":26},{"id":"7db2d0f898a35bd6172bcc72d62361a02d3c7b76","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-21 14:51:11.000000000","message":"Patch Set 26:\n\nUNSTABLE build log link http://tempest-openstack.spb.yadro.com/gerrit-test/170","accounts_in_message":[],"_revision_number":26},{"id":"b82e720b75180fa268f9c24223cda66ec96fad52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-21 16:49:18.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"a546af8c4b8a51d92dda65d2022bc9a63c233a0d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-21 16:51:50.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":27},{"id":"0fbe16ee0f2ef41f394b190226c8b2748d37504b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-21 17:04:57.000000000","message":"Patch Set 26:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-test/170/ : ABORTED\n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-trigger/200/ : ABORTED","accounts_in_message":[],"_revision_number":26},{"id":"62a01158cfdd5e8e3c66906b71dd2c298d1efc9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 17:20:18.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3309ca383034a6392971577da810b34\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a95abe0bc95a433fa04f5ea734e7344d : SUCCESS in 29m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9542359ac474a0daf0a9314f913f8a0 : FAILURE in 22m 43s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"a24cf4587657e9379ca20b229330c2e0f72c998b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-21 18:21:17.000000000","message":"Patch Set 27:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/92/825492/27/check/fujitsu-eternus-dx-iscsi/1c82eac : SUCCESS in 1h 31m 19s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/92/825492/27/check/fujitsu-eternus-dx-fc/e78cddc : SUCCESS in 1h 29m 57s","accounts_in_message":[],"_revision_number":27},{"id":"e61162c86877c03bb6ff8366abb33925f949cd65","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-21 18:21:19.000000000","message":"Patch Set 27:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/27/iscsi/ : FAILURE in 1h 31m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/27/smb/ : SUCCESS in 1h 09m 40s","accounts_in_message":[],"_revision_number":27},{"id":"29df128866a477a8ca1118bbb87904f060b3f0d3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-21 18:22:22.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/92/825492/27/check/iscsi-driver-alletra-6k/84b831c/84b831c.html : FAILURE in 1h 32m 00s","accounts_in_message":[],"_revision_number":27},{"id":"78fc88af808c4151c1808160569499af14724025","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-21 18:30:49.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/825492 : SUCCESS in 1h 40m 04s","accounts_in_message":[],"_revision_number":27},{"id":"93f8ab68aea802322001ade91380e12b80525cb7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-21 18:32:49.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/27/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/64a767a/ : SUCCESS in 1h 25m 08s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/825492/27/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d9fe912/ : SUCCESS in 1h 42m 11s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/92/825492/27/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/b1e69eb/ : FAILURE in 32m 55s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"a8f2e051e009d6628a527682a47cbbe54238d432","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-21 19:02:24.000000000","message":"Patch Set 27:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/825492/27/check/devstack-plugin-ceph-dsvm-focal-py3/57937ba/ : SUCCESS in 2h 08m 24s","accounts_in_message":[],"_revision_number":27},{"id":"e3523222266b16ce0547b555c17e5542e6308ac6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 19:17:55.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbe38e23e4a246639b7adf2518016cec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81ac332509bf4bafad78c71a938ad0d0 : SUCCESS in 8m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/488250916dae449daeb323a06a7b39f4 : SUCCESS in 11m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0d2b489c6144858958ae1bdd46cca76 : SUCCESS in 12m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ffa17df5e1b47e79246b3f3e6a06370 : FAILURE in 12m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78b2644b426546ac979bfb96b1e2586a : SUCCESS in 13m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/f72469da61c445a390beaa0d01813272 : SUCCESS in 58m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2612fbccc23648418691a880f13db5f2 : SUCCESS in 1h 21m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ea3d8e5ff6646748090b794d7284f3e : SUCCESS in 50m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b3590e44a5e49dcaa10363448058842 : SUCCESS in 9m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ebfd564d37e34469bdeb265b4d89cb76 : SUCCESS in 16m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/07fd72279f924cc7bcc3b9b033bcbf35 : FAILURE in 6m 31s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/28395e65c21247728796e26864cf591d : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0020d986f8a148e192a06532b6865b2a : SUCCESS in 7m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2a1300943d084a3eb6f3dbe1d4cc0f70 : SUCCESS in 7m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e425ce16bce2484eb94a5fcbb4b98167 : SUCCESS in 39m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cef3672639bf49538850eb544f4d42da : FAILURE in 7m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8f18e59091fc48109f88dfbde6af2b94 : SUCCESS in 1h 43m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/607def38c96d40e09e2933f2a4a0bd62 : FAILURE in 1h 41m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f5fd108cc9c24c06bb513bff15636ea4 : SUCCESS in 1h 37m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/08b879f1447e4f18b076141d5a8063f1 : SUCCESS in 1h 10m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/babb92a647394140a6e96b368f9e014d : SUCCESS in 1h 43m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2eda402f07d74df9af0e4a58e477e868 : SUCCESS in 47m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/578acfa8fc9247d99a65fcc0c5982722 : SUCCESS in 1h 27m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/748f1cdf1748473fba52d1dfb67e5846 : SUCCESS in 2h 23m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b54c075b962a448b8c4b221cec52f9c6 : SUCCESS in 1h 05m 58s","accounts_in_message":[],"_revision_number":27},{"id":"6a28186047332e86abbe6c4e20acaa22ead9231f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-21 19:19:23.000000000","message":"Patch Set 27:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/201","accounts_in_message":[],"_revision_number":27},{"id":"b95fe8a869f89e17ceb0c279689c2832ac29bd4e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-21 20:13:54.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/825492/27/check/iscsi-driver-nimble/a119b8e/a119b8e.html : FAILURE in 1h 43m 22s","accounts_in_message":[],"_revision_number":27},{"id":"40ae073446bc8b41cc5a38ebbdc266153a68e5d7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-21 20:21:23.000000000","message":"Patch Set 27:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-27/ns5_nfs : SUCCESS in 3h 23m 52s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-27/ns5_iscsi : SUCCESS in 3h 31m 35s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":27},{"id":"04fc622741ad324aa1b6fd0ef2b448d655360463","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-21 20:26:10.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/27/upstream-check/cinder-tempest-plugin-ontap-iscsi/c7d46de/ : SUCCESS in 1h 07m 17s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/27/upstream-check/cinder-tempest-plugin-ontap-nfs/9f02bcc/ : RETRY_LIMIT in 8m 35s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":27},{"id":"a2920d12dc0deb287f1a7dba459431d31db8a446","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-21 20:48:00.000000000","message":"Patch Set 27:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/825492/27/check/DellEMC_Unity_FC/82cb67f/DellEMC_Unity_FC/1221 : SUCCESS in 1h 14m 16s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/27/check/DellEMC_Unity_iSCSI/a1e10cd/DellEMC_Unity_iSCSI/1125 : SUCCESS in 1h 44m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":27},{"id":"366d8895da29b4a90bc2f2ce2ea571e9a96c894c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-21 20:51:18.000000000","message":"Patch Set 27:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/92/825492/27/check/DellEMC_PowerFlex/d722ecb/DellEMC_PowerFlex/1046 : FAILURE in 1h 33m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":27},{"id":"784605abbdc062c9f5b1e9c567f2221ad76d11b6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-21 22:11:58.000000000","message":"Patch Set 27:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/92/825492/27/check/DellEMC_PowerStore_NVMeTCP/f41ed38/DellEMC_PowerStore_NVMeTCP/24 : FAILURE in 2h 53m 42s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":27},{"id":"57c1a4f597474ccd5fbb6a13833810b46007a5c6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-22 00:26:19.000000000","message":"Patch Set 27:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/206","accounts_in_message":[],"_revision_number":27},{"id":"26fbf2c2e5e371809e23ce64fe6085a608e69f56","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-22 06:55:45.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"cb72003847b2ef7974238b6f1ea3dd7b843b1b21","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-22 10:17:52.000000000","message":"Patch Set 27:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/216","accounts_in_message":[],"_revision_number":27},{"id":"ac0c1dbcdf051c43c4809728448c7b063bfa5288","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-07-24 23:49:51.000000000","message":"Patch Set 27: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":27},{"id":"9c0b34d805241ccc0c2a711e221b0c7f5b64d5be","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-25 01:47:19.000000000","message":"Patch Set 27:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":27},{"id":"f694397e458ff1ccb39a449e2d45c26bc5f135b8","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-25 07:14:07.000000000","message":"Patch Set 27:\n\n(2 comments)","accounts_in_message":[],"_revision_number":27},{"id":"04fb92c517d435b5b150641166feae372066361c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-25 10:25:40.000000000","message":"Patch Set 27:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/234","accounts_in_message":[],"_revision_number":27},{"id":"9ef901133a8782f51ca00e628e45cfc3df7cbcdf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-25 10:43:55.000000000","message":"Patch Set 27:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/235","accounts_in_message":[],"_revision_number":27},{"id":"3a5c0c46c07d69c3dabb9f25f9090b5837d79676","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-25 11:54:58.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"d4d42fafc8cdbebc4b4685fb9d73c003b4c46d40","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-25 12:00:05.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"fc6fc0b2207c851f887e41f9d3d3cbf48c302118","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-25 12:00:40.000000000","message":"Patch Set 28:\n\n(2 comments)","accounts_in_message":[],"_revision_number":28},{"id":"af0ae49009d9c3a87ef0b666f224816cd4b74e50","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-25 12:07:31.000000000","message":"Patch Set 28:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/237","accounts_in_message":[],"_revision_number":28},{"id":"ec6f5d6059b75b1a3fb63f020f7fbe7ba148d0e0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-25 12:13:45.000000000","message":"Patch Set 28:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1658750106428 : FAILURE","accounts_in_message":[],"_revision_number":28},{"id":"589cf9ede054a7bfde316fedc0847b7e6f5f954e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 12:16:56.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4eca6cb3f684c45bec589850aa6543d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e98a4d1b069e4a8bab0a0ab4f3aa5415 : SUCCESS in 15m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e955dbed25c44b39d88bce6254a2fa9 : SUCCESS in 15m 58s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"be41c82020decd2675877c332ae8798642397fc4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-25 13:04:42.000000000","message":"Patch Set 28:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/92/825492/28/check/iscsi-driver-alletra-6k/7ac423b/7ac423b.html : FAILURE in 1h 05m 13s","accounts_in_message":[],"_revision_number":28},{"id":"ff398ecd98b61c1c8059a06b649e2948ffd20aee","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-25 13:10:32.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/28/iscsi/ : SUCCESS in 1h 09m 30s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/28/smb/ : SUCCESS in 1h 14m 56s","accounts_in_message":[],"_revision_number":28},{"id":"12b11c001f6c4a8da8da7f268295c555c97156d3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-25 13:22:37.000000000","message":"Patch Set 28:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":28},{"id":"da845c7a0410df6ee9c944b61a6bd125e5045697","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-25 13:29:23.000000000","message":"Patch Set 28:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/92/825492/28/check/fujitsu-eternus-dx-iscsi/12a6fb5 : FAILURE in 1h 19m 29s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/92/825492/28/check/fujitsu-eternus-dx-fc/bcca1da : FAILURE in 1h 33m 45s","accounts_in_message":[],"_revision_number":28},{"id":"3a650b77e5b75d0b2a1af3a843c5ffb415e357a1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-25 13:29:58.000000000","message":"Patch Set 28:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/969f81f422cd4706acb95ce47e317a4f : POST_FAILURE in 1h 27m 11s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/407dcdb8ca464ae48064e8878be7a4ab : POST_FAILURE in 1h 31m 29s","accounts_in_message":[],"_revision_number":28},{"id":"74dc88a1ba701e5bc13cf78e3f5394693c0c1cd0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-25 14:03:48.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/28/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5416c1e/ : SUCCESS in 1h 37m 38s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/825492/28/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/e0c7da6/ : SUCCESS in 1h 44m 46s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/92/825492/28/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/7c7bd3e/ : FAILURE in 28m 19s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"a008f1c3652dbfd75525f43642aeaf3954624ea5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 14:08:43.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44f08f64e69949b3b637225dfd11e05e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbc279f7851f4b08954fdee42a75694a : SUCCESS in 8m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c46428a365594d48a4e3e90089d6e96a : SUCCESS in 12m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1049fd4ccc2d456da893ae234b9adb85 : SUCCESS in 13m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14cf009d487b41b7840ae608225374dc : SUCCESS in 12m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab757b94b7cc46159b16aa147a1801a9 : SUCCESS in 14m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/bcc5322cabcb4ab7aa333ee555778c3b : SUCCESS in 51m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/60895c1ec291453b8a59d473d0913f0a : SUCCESS in 1h 22m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df93661de2d240ed864c8d4ba65ba7d2 : SUCCESS in 43m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50b646b4399141928fe14e18f912424a : SUCCESS in 9m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2f0cfad3b0c94712ab25a64594a49c5f : SUCCESS in 15m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2b5081c6972b44609f31515cb1a1bc12 : FAILURE in 4m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/914d27c84eb040f0a5407160270c0724 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ef8bd0d40e4e47e499b6904a426472f4 : SUCCESS in 5m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eff10bb5acd6400b999264efbc18cb47 : SUCCESS in 7m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e79ee5a35caf4bc8a52c9793529b1705 : SUCCESS in 31m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2c208fe235724e349909b41dc9517bb1 : FAILURE in 7m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/48cc5c218f8e4f8d89cbc454f7c93432 : SUCCESS in 1h 15m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f6b8b83748644b07a674e4c2947acc3c : FAILURE in 1h 28m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eea997b9c2b547dc81a3a20abf4126ed : SUCCESS in 1h 35m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6e917e0bb90a437d8e18d6f386591c45 : SUCCESS in 52m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1a29aad1eb954b498fe9fd883eb18877 : SUCCESS in 1h 22m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/316c351ad414468d84095377c7ab656c : SUCCESS in 36m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2ece6ea52375419182db24e1859e64cf : SUCCESS in 1h 14m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e7424d6ec86c4b56b11fea4160157994 : SUCCESS in 2h 11m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89211ff222cc4aeb947aa8bbf6ba085d : SUCCESS in 44m 33s","accounts_in_message":[],"_revision_number":28},{"id":"4a322786a27357b88176d1ce44c2e016a5459c14","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-25 14:52:07.000000000","message":"Patch Set 28:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/28/upstream-check/cinder-tempest-plugin-ontap-iscsi/dd59169/ : SUCCESS in 42m 04s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/28/upstream-check/cinder-tempest-plugin-ontap-nfs/25a9a9d/ : RETRY_LIMIT in 7m 31s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":28},{"id":"b669960d906ed1d62f57181d3531cd25adee756a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-25 15:21:06.000000000","message":"Patch Set 28:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-28/ns5_nfs : SUCCESS in 3h 25m 37s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-28/ns5_iscsi : SUCCESS in 3h 21m 34s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":28},{"id":"f4046365f25507dbf108367222996a66dc215afe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-25 15:25:59.000000000","message":"Patch Set 28:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/238","accounts_in_message":[],"_revision_number":28},{"id":"fbd1a97b1692d4fcf6490ded1118a54872660780","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-25 16:22:14.000000000","message":"Patch Set 28:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/92/825492/28/check/DellEMC_PowerStore_NFS/56dc799/DellEMC_PowerStore_NFS/167 : FAILURE in 47m 29s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":28},{"id":"a8f1faf925a840b29784d4bd486fb02393c0272c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-25 16:50:08.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"eb25c155f9962f24897a3ee9924d0680eca21a79","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-25 16:50:21.000000000","message":"Patch Set 28:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-trigger/241/ : ABORTED","accounts_in_message":[],"_revision_number":28},{"id":"ef9023d473cadbd6219bd00a2bd81eeeaffabc07","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-25 16:53:27.000000000","message":"Patch Set 29:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/92/825492/29/check/fujitsu-eternus-dx-iscsi/8e04293 : FAILURE in 2m 35s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/92/825492/29/check/fujitsu-eternus-dx-fc/705792f : FAILURE in 2m 35s","accounts_in_message":[],"_revision_number":29},{"id":"a8713570f4990bdded4b80064fbe7f6c0e8c950a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-25 17:08:42.000000000","message":"Patch Set 29:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1658767816959 : FAILURE","accounts_in_message":[],"_revision_number":29},{"id":"855f223fca2fd3d60432065a8f73ba9ab2019629","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 17:38:12.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4e71b58bcf4404b86015519c8bb7001\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c95f3f6ccb14671b15e282abf33550f : SUCCESS in 25m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a8564198a9c6409f92abac09fb357be8 : FAILURE in 41m 08s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"29d7bdacfc89e7b301d12e7d4ac67edfeb86e000","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-25 17:54:11.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/92/825492/29/check/iscsi-driver-alletra-6k/aabea43/aabea43.html : FAILURE in 1h 02m 23s","accounts_in_message":[],"_revision_number":29},{"id":"af09895a1320f10b66d66b96b7a362ff4166afe4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-25 18:19:07.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/825492 : SUCCESS in 1h 24m 11s","accounts_in_message":[],"_revision_number":29},{"id":"e3fbd49d98a3e37903d32f4a792d2e97287edc06","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-25 18:30:33.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/29/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/3e88af2/ : SUCCESS in 1h 39m 23s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/825492/29/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/bdd00c8/ : SUCCESS in 1h 31m 35s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/92/825492/29/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/e30f034/ : FAILURE in 32m 11s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"8f6c680877dd7898824f2049a0d36f8b3674ea2d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-25 19:04:54.000000000","message":"Patch Set 29:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/29/iscsi/ : FAILURE in 1h 12m 18s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/29/smb/ : SUCCESS in 1h 16m 18s","accounts_in_message":[],"_revision_number":29},{"id":"c472261010267961a87dacf3d87d3c5ec0253f55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 19:05:52.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89004bb1b58c45079be81dbac41d515c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d6d67a958974044884f267df069a5c8 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/240387cbd10e4600b6553cd512fdfa51 : SUCCESS in 12m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79a9bc39047c4a9d865e3547f3e5c4dc : SUCCESS in 12m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5837854e13994854a207bcc930a4b772 : SUCCESS in 12m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/077f9951f7b549ed82883e6cf42c4a3b : SUCCESS in 12m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/cb719c92a65f47749c78777652429b18 : SUCCESS in 55m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/69604662b793476fa42b0330a8689c64 : SUCCESS in 1h 17m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4801a52280148f488e372aa74fd4df8 : SUCCESS in 49m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/868a3e5a54eb440da533e6b9e9317d2b : SUCCESS in 8m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3ee373a64e1e43dba6daa3bdcc6b3702 : SUCCESS in 18m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/051554106c3b4695ab211c2ebc95ebc3 : FAILURE in 5m 48s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f439d069505b4ad69bcd04875ebe9669 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/41f78a4ae13144249bfd79b8f77f0ce1 : SUCCESS in 6m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d2e0ee6227c242f382a94f330c7660ed : SUCCESS in 7m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/390ef149197d4095b38a6b09be5a2bf5 : SUCCESS in 37m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b11fa2678c584a329d5e0a841d628534 : FAILURE in 7m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cc3bb1f01f4f48c895793161d2ee720a : SUCCESS in 1h 37m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f46e940087ca4edfab618bb4359f14a1 : FAILURE in 1h 27m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a4f41abdb2be479abacb31639c8c0e5f : SUCCESS in 2h 05m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e24704c7e1724154ac73b6c908d16d11 : SUCCESS in 1h 18m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6874b215ec3b402fbf01781ff6c53de0 : SUCCESS in 1h 42m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ba428ea12d854a54a49f4edf7a0e2461 : SUCCESS in 36m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/698f8cd6b4ed48729150d9e489b5e8bf : SUCCESS in 1h 29m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d3e3b82bd1cc40c29669e51fce584023 : SUCCESS in 2h 13m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43441b8fdadf4720a7035ae0ca726fb7 : SUCCESS in 51m 31s","accounts_in_message":[],"_revision_number":29},{"id":"23b79aab3b8b2f878e764bfe6045451b6980ff3e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-07-25 19:22:11.000000000","message":"Patch Set 29:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/25_07_2022_19_14_23\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":29},{"id":"c73b45f9c39784237f6866d807487d4e462eddc3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-25 19:49:46.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/29/upstream-check/cinder-tempest-plugin-ontap-iscsi/68e8b09/ : SUCCESS in 42m 13s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/29/upstream-check/cinder-tempest-plugin-ontap-nfs/03a5192/ : RETRY_LIMIT in 8m 11s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":29},{"id":"dda40fbd97236181423d964c64d03edbbb7277f6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-25 20:08:30.000000000","message":"Patch Set 29:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/242","accounts_in_message":[],"_revision_number":29},{"id":"623236a3cf1e1b1dda9acabc179e75a516c5cb68","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-25 20:45:57.000000000","message":"Patch Set 29:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/92/825492/29/check/DellEMC_PowerStore_NFS/c40de41/DellEMC_PowerStore_NFS/171 : FAILURE in 42m 38s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":29},{"id":"bce4cd2b0d14f8a363736f481ffb40605e012b3e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-25 20:47:18.000000000","message":"Patch Set 29:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-trigger/258/ : ABORTED","accounts_in_message":[],"_revision_number":29},{"id":"c484ed27c0b7b49694ca3ebcfd101fdf09fc3f9b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-25 20:53:27.000000000","message":"Patch Set 29:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-29/ns5_nfs : FAILURE in 4h 2m 48s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-29/ns5_iscsi : SUCCESS in 3h 29m 23s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":29},{"id":"fbaf7de271a5505145ec287e103edeedf1b59078","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-25 20:54:26.000000000","message":"Patch Set 29:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/259","accounts_in_message":[],"_revision_number":29},{"id":"88dd0785de365d16b946521b4b5d5f9ed8bb81b8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-25 21:48:39.000000000","message":"Patch Set 29:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/92/825492/29/check/DellEMC_PowerFlex/1e51553/DellEMC_PowerFlex/1069 : SUCCESS in 1h 24m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":29},{"id":"460eb93942876f91498c474dcafcf0b8440fc87d","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-07-25 21:49:56.000000000","message":"Patch Set 29:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/92/825492/29/check/DellEMC_PowerMAX_FC/df54a78/DellEMC_PowerMAX_FC/749 : FAILURE in 43m 55s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/29/check/DellEMC_PowerMAX_iSCSI/cd6607c/DellEMC_PowerMAX_iSCSI/733 : FAILURE in 47m 33s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":29},{"id":"a87baf29420d9f40f4cbbf71cda0c8fcaf5ca1a1","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-07-26 00:21:00.000000000","message":"Patch Set 29:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/92/825492/29/check/DellEMC_SC_FC/cafe417/DellEMC_SC_FC/802 : SUCCESS in 53m 36s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/29/check/DellEMC_SC_iSCSI/d699271/DellEMC_SC_iSCSI/824 : SUCCESS in 2h 00m 33s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":29},{"id":"413132334c79ffd2466400544725abc730399a29","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-26 02:18:54.000000000","message":"Patch Set 29:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/825492/29/check/DellEMC_Unity_FC/bfb2068/DellEMC_Unity_FC/1251 : FAILURE in 20m 58s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/29/check/DellEMC_Unity_iSCSI/3e88f6c/DellEMC_Unity_iSCSI/1154 : SUCCESS in 1h 51m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":29},{"id":"c364c552c62be64ff89e997a4ba6a540eed15466","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-26 05:24:38.000000000","message":"Patch Set 28:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":28},{"id":"543024c6e4f0bae2329e1d979acf5b32028e54b1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-26 05:24:42.000000000","message":"Patch Set 29:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":29},{"id":"85c42779a1c93555b91ab7c1f15e59b11ee75a78","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-26 12:54:30.000000000","message":"Patch Set 29:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/260","accounts_in_message":[],"_revision_number":29},{"id":"a42ebea8bf1690255921ed0cd5f18b1abef07513","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-26 12:57:25.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"2ac5fef03cc979e0371e5436519cd9a2c6a3a208","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-26 12:58:54.000000000","message":"Patch Set 30:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":30},{"id":"e56f60d5d4a1bd5c633fc9f1527ff1685cfc7db0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-26 13:15:44.000000000","message":"Patch Set 30:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1658840254568 : FAILURE","accounts_in_message":[],"_revision_number":30},{"id":"2df3651a63672995f4f3a0dc512a3fb16c81ac2d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-26 13:37:46.000000000","message":"Patch Set 30:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/263","accounts_in_message":[],"_revision_number":30},{"id":"1ba44483b5543395804f5c1f4184d38fd54e3fa6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-26 14:01:18.000000000","message":"Patch Set 30:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/92/825492/30/check/iscsi-driver-alletra-6k/4396ab5/4396ab5.html : FAILURE in 1h 02m 51s","accounts_in_message":[],"_revision_number":30},{"id":"cf8ceb93b00b762edfa6b67218338c9f71a9984a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 14:16:34.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/428a6a4db8df471997d6c7cbf046cd7f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/199eeaadaa994e0dbf49d9526e192737 : SUCCESS in 25m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/322c1cfd52814bf78621de563c58f12f : SUCCESS in 23m 17s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"e2abf45248efe02bae8a18246e492a586d070d9b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-26 14:30:06.000000000","message":"Patch Set 30:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/825492 : FAILURE in 1h 27m 42s","accounts_in_message":[],"_revision_number":30},{"id":"bcce396bc45c800e303c1f0e8ed91a9693804e0f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-26 14:30:39.000000000","message":"Patch Set 30:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/30/iscsi/ : FAILURE in 1h 32m 37s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/30/smb/ : SUCCESS in 1h 05m 52s","accounts_in_message":[],"_revision_number":30},{"id":"32557981a5fb5474246e82ed053614dd36422ba9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-26 14:34:36.000000000","message":"Patch Set 30:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/30/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c1b71b9/ : SUCCESS in 1h 25m 36s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/825492/30/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b381f40/ : SUCCESS in 1h 36m 10s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/92/825492/30/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/a8b2ca7/ : FAILURE in 27m 44s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"93b7777dec5c5a2fe3eb3011d86cc3937459d377","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-26 14:50:52.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"968b62ac91528ff13d50e5232357d8d809818003","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-26 14:50:56.000000000","message":"Patch Set 30:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-trigger/268/ : ABORTED","accounts_in_message":[],"_revision_number":30},{"id":"110a98827a5c5fb617bf5b9913eccf815f62faa5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-26 14:56:14.000000000","message":"Patch Set 30:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/f3935a73ee0b45baa39d721f0382868e : POST_FAILURE in 1h 50m 18s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/7e61bad5ba954a559d9aacae595fa922 : POST_FAILURE in 1h 55m 11s","accounts_in_message":[],"_revision_number":30},{"id":"4ce42033506353eed072e46252947f8412e7933a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 15:27:42.000000000","message":"Patch Set 30: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bbc1451c253c4c5189f78eb2d49261f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19a32231220147c6881e5d98733467f6 : FAILURE in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89d34eb00dd640d28efc051719136e22 : SUCCESS in 10m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/314eef6828d94b39a67370c8be418d16 : SUCCESS in 10m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a843ba7df5e94d7aa3f106d5327275c9 : FAILURE in 11m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97be1fa8f4f54a22a396ce8de1d60ae9 : SUCCESS in 10m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/6e2b207f70e04c10a32e4374d56f3a11 : SUCCESS in 1h 04m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8947c00a7696402eaf256027a555b593 : SUCCESS in 2h 03m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1283516109641429f69f4f2f1886737 : SUCCESS in 54m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e99e7edfb8b4166bf566314010cad0a : SUCCESS in 8m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/484c9f22dfd2464c9effd9e69d258632 : SUCCESS in 15m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9de44179b8494424bd616d2a1774e818 : FAILURE in 4m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b3944298273444c08abe965553dbe054 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/16f32655f37b48fb8eb87b135f3045ee : SUCCESS in 6m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5664860f90034a32b41be4322e4e1f10 : SUCCESS in 5m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/58c776d9f5fa47c7984dd8d550c254e1 : SUCCESS in 46m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2f4e6f37f6654695b88b3d31c00d3a8f : FAILURE in 5m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e6381f406c75458c8d9e2a20518dc2d8 : SUCCESS in 1h 37m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/05c914ae42d4434980238c7f8ae55ce8 : FAILURE in 1h 31m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b634c8dc2e7e4a3e83c1a8bace143989 : SUCCESS in 2h 02m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/88555cbd3eaa4f9797bcc383c37a503e : SUCCESS in 1h 16m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4c5473cba2924234bcc657ee356b0b5e : SUCCESS in 1h 29m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8090669a24b04309a0697d23a64d8f0a : SUCCESS in 44m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0d61b53288074760aa3c185a5f126db4 : SUCCESS in 1h 22m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/87a38a5327094ab69a058a84e0242497 : SUCCESS in 2h 29m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e4576dc42ed4140bf3444e408450f14 : SUCCESS in 52m 00s","accounts_in_message":[],"_revision_number":30},{"id":"c4e31f50a21d0b96a32c494ff6411ca213d5c9d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-26 17:28:31.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"96addecfee73de4b0dc922864dec936ec97715c2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-26 17:28:34.000000000","message":"Patch Set 30:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-trigger/271/ : ABORTED","accounts_in_message":[],"_revision_number":30},{"id":"85afa1adf8bbd2fbac1cf767a868f6f234605280","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-26 17:30:28.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":31},{"id":"2ef25dc827c7575dda71e396de4cdc015c42ec29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:59:52.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91de0103747d49a3ab7453c6a88a209a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/806035976eaa488ca3d55aa159151bd9 : SUCCESS in 28m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b8f13b5b7f145f1b749de0515e741fe : SUCCESS in 24m 59s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"9be2a955f5c67d0f0b8288ca9219b8ee46a97693","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-26 18:33:11.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/92/825492/31/check/iscsi-driver-alletra-6k/0d7d943/0d7d943.html : FAILURE in 1h 02m 53s","accounts_in_message":[],"_revision_number":31},{"id":"6a597aefa4ed0d86be8b7ae99e3cfec74b82c809","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-26 18:34:39.000000000","message":"Patch Set 31:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1658856520470 : FAILURE","accounts_in_message":[],"_revision_number":31},{"id":"41bd6bbc113f005607148ada96246f5aada82193","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-26 19:13:39.000000000","message":"Patch Set 31:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/31/iscsi/ : FAILURE in 1h 44m 31s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/31/smb/ : SUCCESS in 1h 04m 14s","accounts_in_message":[],"_revision_number":31},{"id":"db2b8d6f0f8ab36d10dcebf876397a1b4372c5fd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-26 19:19:44.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/825492 : SUCCESS in 1h 38m 23s","accounts_in_message":[],"_revision_number":31},{"id":"c761d96fd81526cc7b39f89349282ba190728582","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 19:31:03.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b85aafacea154fbcbae648a220c75196\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e94d927e7fe549be9e7e89ae79da1c5c : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23ab868b958b4c95b434f0f960a82881 : SUCCESS in 9m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43484583cd5a4658a325cad064115248 : SUCCESS in 14m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a726ea0deab54db98bb1ac7922c7a354 : FAILURE in 11m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3835508c79f64398ab83ecc208ff5c2e : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/724be0c912ea4c4c8184ab98136ab61a : SUCCESS in 1h 11m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b1b59da16dbb41af88957fdcb8a7951c : SUCCESS in 1h 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8dfbcc7587234d9d980f4b948fc38402 : SUCCESS in 1h 03m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9c6b07f86b940d2ac85a0b1213ea15b : SUCCESS in 7m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/687e75a84b8d4dd1bd8432175a239a71 : SUCCESS in 14m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/232f509e159a44799d4a0f1e785ebb79 : FAILURE in 4m 36s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d7d103a45d7e484499fdb57ae22c9614 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8c4348afba634636b7950c194d64258f : SUCCESS in 6m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0ff00f6ebf93414d9a58e1e8aaef34f1 : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2daf603006404936a8708878391ac8c9 : SUCCESS in 1h 01m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cd3471e67c5a4fbd825985da9c5211e1 : FAILURE in 6m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3066f541252a4a319e30d182ae75f396 : SUCCESS in 1h 17m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/72ab7dc32b0d4a1da5f8143d1cfd8702 : FAILURE in 1h 51m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0750bf6bf6594128a0159bd0f8e24d8e : SUCCESS in 1h 43m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f75a0f66ca3d40c5b98e57d257383124 : SUCCESS in 59m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a4eb7383b37246bd9c1d8182edfe35b0 : SUCCESS in 1h 20m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f4c6ec44af2d47c0a2dc9e17de09a187 : SUCCESS in 46m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6bd2fc9a53a04ff2a4535c0523440378 : SUCCESS in 1h 42m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d646a930391e4856a43d183264d581d2 : SUCCESS in 2h 00m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a98c9c71be364c2e8f472213f5e0a353 : SUCCESS in 49m 27s","accounts_in_message":[],"_revision_number":31},{"id":"a37b70aa16b7bbb82cd5fdde337441a565ea93c2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-07-26 19:44:58.000000000","message":"Patch Set 31:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/26_07_2022_19_37_07\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":31},{"id":"991d167cdc013a8670cda2476a9b87d30c049dad","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-07-26 20:02:08.000000000","message":"Patch Set 31: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":31},{"id":"be4bad60281345a41f4b834f5ecde8cb15f70e7c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-26 20:16:42.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp finger://managesf.services.openstack.netapp.com:7979/b2828be6842c45718143467ad668b1b5 : RETRY_LIMIT in 1m 25s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/31/upstream-check/cinder-tempest-plugin-ontap-iscsi/089d679/ : FAILURE in 44m 28s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/31/upstream-check/cinder-tempest-plugin-ontap-nfs/1862dbd/ : SUCCESS in 41m 24s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":31},{"id":"75a93fd24d8c1411af954feb0acc981f31ca8a76","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-26 20:36:32.000000000","message":"Patch Set 31:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/666ebcaadc4c4bd284bec4da9b73388a : POST_FAILURE in 1h 59m 22s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/f4b22be370e84348aa8f73a2e2b0d9d2 : POST_FAILURE in 3h 00m 16s","accounts_in_message":[],"_revision_number":31},{"id":"20f9d234b5730e1f3e3e5e4baa6dd954ecf21da3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-26 20:57:45.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/31/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/92bbe8d/ : SUCCESS in 1h 38m 18s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/825492/31/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/8f78c44/ : SUCCESS in 1h 35m 15s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/92/825492/31/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/303d962/ : FAILURE in 31m 13s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"fead9692b3c0c2733872d9193a963fcf8e12d41f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-26 21:31:25.000000000","message":"Patch Set 31:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/92/825492/31/check/DellEMC_PowerStore_NFS/15ff8e9/DellEMC_PowerStore_NFS/183 : FAILURE in 48m 40s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":31},{"id":"96c820970e2aedfcc373a5ebd31f74a4bf164c4c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-26 22:15:34.000000000","message":"Patch Set 31:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/275","accounts_in_message":[],"_revision_number":31},{"id":"b16489a77112b851bdae94273e557f5715e0cf9e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-27 01:38:09.000000000","message":"Patch Set 31:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/92/825492/31/check/DellEMC_PowerFlex/b163400/DellEMC_PowerFlex/1081 : SUCCESS in 1h 24m 48s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":31},{"id":"9d4822afa4dacd9b70dd08827db68cab5a5d6121","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-27 02:14:22.000000000","message":"Patch Set 31:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":31},{"id":"4921d1d15efbb96f9eebdb5fa0bff40c6d88cbc2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-27 02:14:29.000000000","message":"Patch Set 30:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":30},{"id":"b268bc93766c69f207efa1a3a1b296a2fbbb3717","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-27 07:06:03.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"f68e6ab66183bd251a3fc2d2a9e770c563ed2760","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-27 07:08:03.000000000","message":"Patch Set 32:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":32},{"id":"f1d3a8ce40556de361faca4378320b11e9ef86db","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-27 07:08:23.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"ffd779c61df45fbc4bc655d0847c183d15591adb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 07:24:53.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7766ffd525e4016bd48ee91da7f2c2c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/049cc6c0a36940d2849f28688386ed06 : SUCCESS in 13m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e96da5eb37a24453b3c7cdf872a4d1cd : SUCCESS in 14m 10s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"2373fed18c0cb699fbcf6e185191d3d9ead084d1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-27 07:35:47.000000000","message":"Patch Set 32:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/92/825492/32/check/iscsi-driver-alletra-6k/d7958a2/d7958a2.html : FAILURE in 27m 30s","accounts_in_message":[],"_revision_number":32},{"id":"c111532f671bc6596e7fdaa58a9a0c1dac5473ac","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-27 07:40:33.000000000","message":"Patch Set 32:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1658905572031 : FAILURE","accounts_in_message":[],"_revision_number":32},{"id":"013bf9b022a593823abd8d2d53d80ffd5b844144","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-27 08:08:29.000000000","message":"Patch Set 32:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/825492/32/iscsi/ : SUCCESS in 57m 45s\n- cinder-smb http://cloudbase-ci.com/cinder/825492/32/smb/ : SUCCESS in 1h 01m 49s","accounts_in_message":[],"_revision_number":32},{"id":"4e3268ccceb02ca250ca9e8222875284e1f83cc5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-27 08:30:58.000000000","message":"Patch Set 30:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-30/ns5_nfs : FAILURE in 19h 33m 1s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-30/ns5_iscsi : FAILURE in 19h 33m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":30},{"id":"12fedfbe83c17bd03af559eaf46f4862111ea633","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-27 08:33:41.000000000","message":"Patch Set 32:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/825492/32/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/80fa327/ : SUCCESS in 1h 24m 28s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/825492/32/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/e34f6ff/ : SUCCESS in 1h 26m 11s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/92/825492/32/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/3055f6e/ : FAILURE in 35m 16s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"58ee342b03399f14b7ca589e944fe3b1865c422c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-27 08:34:26.000000000","message":"Patch Set 32:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":32},{"id":"9ca00ee8d44119f4de1331daff4f0ffd38c9821d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-27 08:48:30.000000000","message":"Patch Set 32:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/825492 : SUCCESS in 1h 39m 12s","accounts_in_message":[],"_revision_number":32},{"id":"51d6cbb125b90f9025cad893d9fa2217c1474a87","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-27 09:06:15.000000000","message":"Patch Set 32:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/7eb2d4f5f3304a34a6baa646a5c43ddb : POST_FAILURE in 1h 31m 23s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/e11557111c4d406f9c10e20ddeef2070 : POST_FAILURE in 1h 35m 11s","accounts_in_message":[],"_revision_number":32},{"id":"0f9fba2ae8955270c4db5522c4057141326a18c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 09:07:26.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1c3e3415c7846f98016955561aefdf3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97b5e76b78ec4fe38b2b368a2f53bbd0 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/192fbac689a74de1aedb8f4cfaf2a59f : SUCCESS in 11m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/316f67ca9c734b0698fca246a7822338 : SUCCESS in 10m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84176532317244508ef48917d74df9f5 : SUCCESS in 11m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbfdbbe75f0645b6a729549019c82a8a : SUCCESS in 10m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/ba63e63eb3464c778d8033a6a38407b2 : SUCCESS in 52m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6985f7f592ca43bfaaab9b72c34435d4 : SUCCESS in 1h 43m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f429482257234fa881fca3ce289d5cd5 : SUCCESS in 43m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0d555b3a2cb45fea9b122a145d51a20 : SUCCESS in 7m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/18f984b13953496f9830a1d35e1fcb2d : SUCCESS in 13m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c704a33c4d4a43719a9222a541cde483 : FAILURE in 5m 41s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/51c1928385e242d7a67eff174f3c2b1c : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a4ebbf3d7a1448fc8d461d5dc1d45c6a : SUCCESS in 6m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ef6fab06977a494e8621c7f2bd4fd1b2 : SUCCESS in 7m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a42f7e97a20b41618ec2b4fe3ba8dbaa : SUCCESS in 33m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d69dbc00b1104677b56b374d8ece6581 : FAILURE in 6m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c096c63151bd4e0488066f98d7fd0836 : SUCCESS in 1h 30m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/563180fc7428476ea60d1c44ee595587 : FAILURE in 1h 12m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1722d20a21dd4b84b2bbf4f6e9013704 : SUCCESS in 1h 59m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ff62072cd7664ddabb74a6b7399327d8 : SUCCESS in 1h 20m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/be6765cdd26f4e1598f710566348f0a2 : SUCCESS in 1h 48m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/19ef474500d7419da7f75e2cde5562a8 : SUCCESS in 54m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/21c067536468466fab10223d3971b801 : SUCCESS in 1h 17m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1a5671e8fa954402afa62ececa5b7e45 : SUCCESS in 1h 57m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4b2ad83840344429036fc396b105a1e : SUCCESS in 58m 36s","accounts_in_message":[],"_revision_number":32},{"id":"5995794494e24e8169779df5ea3befa50bbc4a59","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-07-27 09:15:42.000000000","message":"Patch Set 32:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/27_07_2022_09_07_35\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":32},{"id":"4ef7dfb05139d9b676e1be0e31244698e562b1b2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-27 10:20:45.000000000","message":"Patch Set 32:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/32/upstream-check/cinder-tempest-plugin-ontap-fcp/14c175d/ : RETRY_LIMIT in 10m 46s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/32/upstream-check/cinder-tempest-plugin-ontap-iscsi/e1288f7/ : FAILURE in 1h 12m 06s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/825492/32/upstream-check/cinder-tempest-plugin-ontap-nfs/bf4d9ea/ : SUCCESS in 42m 24s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":32},{"id":"03751e34da42b5fc823bf18e2b3a569703146958","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-27 10:31:36.000000000","message":"Patch Set 32:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/281","accounts_in_message":[],"_revision_number":32},{"id":"55abfd9aef9808abf1f0890daea6d286b9dc7a3a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-27 10:54:24.000000000","message":"Patch Set 32:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/92/825492/32/check/DellEMC_PowerFlex/706b068/DellEMC_PowerFlex/1083 : SUCCESS in 1h 28m 28s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":32},{"id":"84548c0665d18b3d30e0e961e3653c80475858d8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-07-27 11:00:57.000000000","message":"Patch Set 32: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"95ac7490f26cde51f06ef7e8c060bbe04cbea07b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-27 12:12:43.000000000","message":"Patch Set 32:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-825492-32/ns5_nfs : SUCCESS in 3h 27m 52s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-825492-32/ns5_iscsi : SUCCESS in 3h 31m 7s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":32},{"id":"752735c9158c43b00283232ea8cb73fbce2ade39","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-07-27 22:27:50.000000000","message":"Patch Set 32:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/92/825492/32/check/DellEMC_SC_FC/42f294f/DellEMC_SC_FC/815 : SUCCESS in 51m 02s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/32/check/DellEMC_SC_iSCSI/147c291/DellEMC_SC_iSCSI/836 : SUCCESS in 1h 57m 36s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":32},{"id":"8884c098c1e6f62bdaf4663b769af66feb1f1222","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-28 03:44:40.000000000","message":"Patch Set 32:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/92/825492/32/check/DellEMC_Unity_FC/75ec7ab/DellEMC_Unity_FC/1274 : FAILURE in 19m 24s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/92/825492/32/check/DellEMC_Unity_iSCSI/ec8e6e2/DellEMC_Unity_iSCSI/1169 : SUCCESS in 1h 52m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":32},{"id":"5e97f2a9e33941b7c99c88bcc6859f3dca2ab85c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-07-28 08:48:14.000000000","message":"Patch Set 32: Code-Review+2 Workflow+1\n\n(14 comments)","accounts_in_message":[],"_revision_number":32},{"id":"284d481e5a07a431637809609a5f22f3b73eecf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 08:48:55.000000000","message":"Patch Set 32: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":32},{"id":"fec91458086d2a2170661e7469922a6c268d620b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 10:26:46.000000000","message":"Patch Set 32: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb5784c9bc144910a8fe6c438ece6724\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f66de63af452447e994c30f468164195 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba59419974f34af482ba1290ff82ea78 : SUCCESS in 12m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c41ce989e53d43d8babaf7ab66b847fa : SUCCESS in 12m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c9a50e0071a431abeb6b88b6af14138 : SUCCESS in 13m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/af721d8a87724b3d8507adc4cb454a09 : SUCCESS in 57m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/10d100e9020646dcb4e36854508db4e7 : SUCCESS in 1h 31m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f4713c9d56b4582862ba6a2ccc001ad : SUCCESS in 53m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12852b5ada3e4d03a965fc74d3ae832f : SUCCESS in 9m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cdd9bae5d8824ed8b684bcaf8fcc68b8 : SUCCESS in 1h 18m 15s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cdced4af3c1b47e6bf327e450a0a5fb3 : SUCCESS in 1h 28m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2a0c2e78e564ec99ffa64a2c8e4580b : SUCCESS in 1h 01m 59s","accounts_in_message":[],"_revision_number":32},{"id":"526ed66affbce35f991071d67c53aa83d1255da7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 10:26:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":32},{"id":"589f235bee73e9df0aad6b15ed940b4e767b4873","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 10:28:21.000000000","message":"Patch Set 32:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7bcbd22aa4841aa8fe267de150070dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9744d01158654039b9d29553f3fb9e82 : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee42165dab324ab78e245d0a4eac90e5 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":32},{"id":"59ca0c7cf1d31f929f1ded9bc7e57cb7a0e079a9","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-07-28 14:45:59.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"cb7671dbf17e45c383cba3cc6221f61a6379e4a8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-28 17:21:05.000000000","message":"Patch Set 32:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/320 \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":32}],"current_revision_number":32,"current_revision":"6934463943f383c5c8d513720caf23f1cef83818","revisions":{"ec58573348ffcc837d41b1e16853868615843ed5":{"kind":"REWORK","_number":1,"created":"2022-01-20 12:30:39.000000000","uploader":{"_account_id":34497,"name":"kid1972","display_name":"Sergey Karpenkov","email":"s.karpenkov@yadro.com","username":"yadro"},"ref":"refs/changes/92/825492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/1"}}},"commit":{"parents":[{"commit":"43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6","subject":"Remove block-box files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6"}]}],"author":{"name":"SKarpenkov","email":"karpenkovs@gmail.com","date":"2022-01-20 12:27:06.000000000","tz":180},"committer":{"name":"SKarpenkov","email":"karpenkovs@gmail.com","date":"2022-01-20 12:29:13.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED cinder driver","message":"Initial commit for Yadro Tatlin.UNIFIED cinder driver\n\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ec58573348ffcc837d41b1e16853868615843ed5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ec58573348ffcc837d41b1e16853868615843ed5"}]},"branch":"refs/heads/master"},"3b2d3a008e9ef990c2bb3bd4e3a5b829987ad5eb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-01-20 13:40:19.000000000","uploader":{"_account_id":34497,"name":"kid1972","display_name":"Sergey Karpenkov","email":"s.karpenkov@yadro.com","username":"yadro"},"ref":"refs/changes/92/825492/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/2"}}},"commit":{"parents":[{"commit":"43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6","subject":"Remove block-box files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6"}]}],"author":{"name":"SKarpenkov","email":"karpenkovs@gmail.com","date":"2022-01-20 12:27:06.000000000","tz":180},"committer":{"name":"SKarpenkov","email":"karpenkovs@gmail.com","date":"2022-01-20 13:39:18.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED cinder driver","message":"Initial commit for Yadro Tatlin.UNIFIED cinder driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3b2d3a008e9ef990c2bb3bd4e3a5b829987ad5eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3b2d3a008e9ef990c2bb3bd4e3a5b829987ad5eb"}]},"branch":"refs/heads/master"},"a43d6f6157092f562ac3a975a5d87644f0bcb794":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-01-20 13:53:54.000000000","uploader":{"_account_id":33437,"name":"YADRO","email":"partners@yadro.com"},"ref":"refs/changes/92/825492/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/3"}}},"commit":{"parents":[{"commit":"43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6","subject":"Remove block-box files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6"}]}],"author":{"name":"SKarpenkov","email":"karpenkovs@gmail.com","date":"2022-01-20 12:27:06.000000000","tz":180},"committer":{"name":"YADRO","email":"partners@yadro.com","date":"2022-01-20 13:53:54.000000000","tz":0},"subject":"Initial commit for Yadro Tatlin.UNIFIED cinder driver","message":"Initial commit for Yadro Tatlin.UNIFIED cinder driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\nImplements: https://blueprints.launchpad.net/cinder/+spec/yadro-tatlin-unified-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a43d6f6157092f562ac3a975a5d87644f0bcb794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a43d6f6157092f562ac3a975a5d87644f0bcb794"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ae80400fd8b6a34342a97f7f25ecf3a45733c25d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-01-20 13:54:59.000000000","uploader":{"_account_id":33437,"name":"YADRO","email":"partners@yadro.com"},"ref":"refs/changes/92/825492/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/4"}}},"commit":{"parents":[{"commit":"43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6","subject":"Remove block-box files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6"}]}],"author":{"name":"SKarpenkov","email":"karpenkovs@gmail.com","date":"2022-01-20 12:27:06.000000000","tz":180},"committer":{"name":"YADRO","email":"partners@yadro.com","date":"2022-01-20 13:54:59.000000000","tz":0},"subject":"Initial commit for Yadro Tatlin.UNIFIED cinder driver","message":"Initial commit for Yadro Tatlin.UNIFIED cinder driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\nImplements: yadro-tatlin-unified-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae80400fd8b6a34342a97f7f25ecf3a45733c25d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae80400fd8b6a34342a97f7f25ecf3a45733c25d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f908887553b84c7ba7cce61bf3c4bb880eb73db6":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-01-20 13:55:28.000000000","uploader":{"_account_id":33437,"name":"YADRO","email":"partners@yadro.com"},"ref":"refs/changes/92/825492/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/5"}}},"commit":{"parents":[{"commit":"43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6","subject":"Remove block-box files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6"}]}],"author":{"name":"SKarpenkov","email":"karpenkovs@gmail.com","date":"2022-01-20 12:27:06.000000000","tz":180},"committer":{"name":"YADRO","email":"partners@yadro.com","date":"2022-01-20 13:55:28.000000000","tz":0},"subject":"Initial commit for Yadro Tatlin.UNIFIED cinder driver","message":"Initial commit for Yadro Tatlin.UNIFIED cinder driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\nImplements: https://blueprints.launchpad.net/cinder/+spec/yadro-tatlin-unified-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f908887553b84c7ba7cce61bf3c4bb880eb73db6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f908887553b84c7ba7cce61bf3c4bb880eb73db6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ccd7460466040a918a3db308f1d0b0997a5a738e":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-01-20 13:57:37.000000000","uploader":{"_account_id":33437,"name":"YADRO","email":"partners@yadro.com"},"ref":"refs/changes/92/825492/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/6"}}},"commit":{"parents":[{"commit":"43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6","subject":"Remove block-box files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6"}]}],"author":{"name":"SKarpenkov","email":"karpenkovs@gmail.com","date":"2022-01-20 12:27:06.000000000","tz":180},"committer":{"name":"YADRO","email":"partners@yadro.com","date":"2022-01-20 13:57:37.000000000","tz":0},"subject":"Initial commit for Yadro Tatlin.UNIFIED cinder driver","message":"Initial commit for Yadro Tatlin.UNIFIED cinder driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\nImplements: yadro-tatlin-unified-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ccd7460466040a918a3db308f1d0b0997a5a738e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ccd7460466040a918a3db308f1d0b0997a5a738e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"531c5896fccf5138a0c60e229731939e5bc3b3a9":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2022-01-20 14:01:16.000000000","uploader":{"_account_id":33437,"name":"YADRO","email":"partners@yadro.com"},"ref":"refs/changes/92/825492/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/7"}}},"commit":{"parents":[{"commit":"43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6","subject":"Remove block-box files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6"}]}],"author":{"name":"SKarpenkov","email":"karpenkovs@gmail.com","date":"2022-01-20 12:27:06.000000000","tz":180},"committer":{"name":"YADRO","email":"partners@yadro.com","date":"2022-01-20 14:01:16.000000000","tz":0},"subject":"Initial commit for Yadro Tatlin.UNIFIED cinder driver","message":"Initial commit for Yadro Tatlin.UNIFIED cinder driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\nImplements: \u003ca href\u003d\"https://blueprints.launchpad.net/cinder/+spec/yadro-tatlin-unified-driver\"\u003eyadro-tatlin-unified-driver\u003c/a\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/531c5896fccf5138a0c60e229731939e5bc3b3a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/531c5896fccf5138a0c60e229731939e5bc3b3a9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"275edcab30153e2be22e52b033a21170c17d8279":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2022-01-20 14:02:34.000000000","uploader":{"_account_id":33437,"name":"YADRO","email":"partners@yadro.com"},"ref":"refs/changes/92/825492/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/8"}}},"commit":{"parents":[{"commit":"43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6","subject":"Remove block-box files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6"}]}],"author":{"name":"SKarpenkov","email":"karpenkovs@gmail.com","date":"2022-01-20 12:27:06.000000000","tz":180},"committer":{"name":"YADRO","email":"partners@yadro.com","date":"2022-01-20 14:02:34.000000000","tz":0},"subject":"Initial commit for Yadro Tatlin.UNIFIED cinder driver","message":"Initial commit for Yadro Tatlin.UNIFIED cinder driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\nImplements: yadro-tatlin-unified-driver\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/275edcab30153e2be22e52b033a21170c17d8279"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/275edcab30153e2be22e52b033a21170c17d8279"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e3d7dcb7581cc4387cb8b802a857862bb88ecd16":{"kind":"REWORK","_number":9,"created":"2022-01-20 15:07:25.000000000","uploader":{"_account_id":34497,"name":"kid1972","display_name":"Sergey Karpenkov","email":"s.karpenkov@yadro.com","username":"yadro"},"ref":"refs/changes/92/825492/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/9"}}},"commit":{"parents":[{"commit":"43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6","subject":"Remove block-box files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6"}]}],"author":{"name":"SKarpenkov","email":"karpenkovs@gmail.com","date":"2022-01-20 12:27:06.000000000","tz":180},"committer":{"name":"SKarpenkov","email":"karpenkovs@gmail.com","date":"2022-01-20 15:07:12.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED cinder driver","message":"Initial commit for Yadro Tatlin.UNIFIED cinder driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3d7dcb7581cc4387cb8b802a857862bb88ecd16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e3d7dcb7581cc4387cb8b802a857862bb88ecd16"}]},"branch":"refs/heads/master"},"5e8ea6f7e8873081b2c7811bb90d4d18073356cb":{"kind":"REWORK","_number":10,"created":"2022-01-20 15:17:07.000000000","uploader":{"_account_id":34497,"name":"kid1972","display_name":"Sergey Karpenkov","email":"s.karpenkov@yadro.com","username":"yadro"},"ref":"refs/changes/92/825492/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/10"}}},"commit":{"parents":[{"commit":"43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6","subject":"Remove block-box files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6"}]}],"author":{"name":"SKarpenkov","email":"karpenkovs@gmail.com","date":"2022-01-20 12:27:06.000000000","tz":180},"committer":{"name":"SKarpenkov","email":"karpenkovs@gmail.com","date":"2022-01-20 15:16:55.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED cinder driver","message":"Initial commit for Yadro Tatlin.UNIFIED cinder driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5e8ea6f7e8873081b2c7811bb90d4d18073356cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5e8ea6f7e8873081b2c7811bb90d4d18073356cb"}]},"branch":"refs/heads/master"},"157a5be2f8236da158e1074b81e516bf32626308":{"kind":"REWORK","_number":11,"created":"2022-05-19 13:56:37.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/11"}}},"commit":{"parents":[{"commit":"43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6","subject":"Remove block-box files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43ca0d6e3c1cf3d98ef8d4f32bfc401bea3d01a6"}]}],"author":{"name":"SKarpenkov","email":"karpenkovs@gmail.com","date":"2022-01-20 12:27:06.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-19 13:56:17.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/157a5be2f8236da158e1074b81e516bf32626308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/157a5be2f8236da158e1074b81e516bf32626308"}]},"branch":"refs/heads/master"},"b4e12001b0168846050353c803e27b16e5f78e96":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2022-05-19 14:03:09.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/12"}}},"commit":{"parents":[{"commit":"cb7f1cc27aa0a706ba25a7f0cb6f65ed731de40b","subject":"Merge \"pylint: ignore __original_module_threading errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cb7f1cc27aa0a706ba25a7f0cb6f65ed731de40b"}]}],"author":{"name":"SKarpenkov","email":"karpenkovs@gmail.com","date":"2022-01-20 12:27:06.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-19 14:03:09.000000000","tz":0},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4e12001b0168846050353c803e27b16e5f78e96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4e12001b0168846050353c803e27b16e5f78e96"}]},"branch":"refs/heads/master","description":"Rebase"},"0197c893e6d43fcd09d07db0888f7fa65d1beec9":{"kind":"REWORK","_number":13,"created":"2022-05-23 12:19:55.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/13"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0197c893e6d43fcd09d07db0888f7fa65d1beec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0197c893e6d43fcd09d07db0888f7fa65d1beec9"}]},"branch":"refs/heads/master"},"e9b1204fdd177558d0c2dc772983e6f4f71b5a24":{"kind":"REWORK","_number":14,"created":"2022-05-23 12:22:33.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/14"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:22:05.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9b1204fdd177558d0c2dc772983e6f4f71b5a24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9b1204fdd177558d0c2dc772983e6f4f71b5a24"}]},"branch":"refs/heads/master"},"04289ee3a7cab0bbeb1af748e5086ef2253661d2":{"kind":"REWORK","_number":15,"created":"2022-05-25 06:36:57.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/15"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-25 06:34:07.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04289ee3a7cab0bbeb1af748e5086ef2253661d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04289ee3a7cab0bbeb1af748e5086ef2253661d2"}]},"branch":"refs/heads/master"},"88cb508ff0bc92c1bb8bef45b1371f63c04ac335":{"kind":"REWORK","_number":16,"created":"2022-05-26 15:49:28.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/16"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-26 15:49:11.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88cb508ff0bc92c1bb8bef45b1371f63c04ac335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/88cb508ff0bc92c1bb8bef45b1371f63c04ac335"}]},"branch":"refs/heads/master"},"1323d6fa190631fba823d0f893e9cb27b4956ccd":{"kind":"REWORK","_number":17,"created":"2022-05-26 15:53:30.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/17"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-26 15:53:15.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1323d6fa190631fba823d0f893e9cb27b4956ccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1323d6fa190631fba823d0f893e9cb27b4956ccd"}]},"branch":"refs/heads/master"},"c0f6cc33ebea1895a91f71d07b75bdc586902aa2":{"kind":"REWORK","_number":18,"created":"2022-05-27 08:34:04.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/18"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-27 08:33:05.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0f6cc33ebea1895a91f71d07b75bdc586902aa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0f6cc33ebea1895a91f71d07b75bdc586902aa2"}]},"branch":"refs/heads/master"},"b5d2abaebceef2d7f2250059fd85acf105278e6d":{"kind":"REWORK","_number":19,"created":"2022-06-02 08:18:52.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/19"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-06-02 08:18:13.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5d2abaebceef2d7f2250059fd85acf105278e6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5d2abaebceef2d7f2250059fd85acf105278e6d"}]},"branch":"refs/heads/master"},"fd629781d9c4e29fb0c16f753a91e2f231f0156d":{"kind":"REWORK","_number":20,"created":"2022-06-06 11:20:10.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/20"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-06-06 11:19:18.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fd629781d9c4e29fb0c16f753a91e2f231f0156d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fd629781d9c4e29fb0c16f753a91e2f231f0156d"}]},"branch":"refs/heads/master"},"3a96d64a6387d4546cd8667fb7d18562628b2d4e":{"kind":"REWORK","_number":21,"created":"2022-06-09 13:04:23.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/21"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-06-09 13:01:01.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3a96d64a6387d4546cd8667fb7d18562628b2d4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3a96d64a6387d4546cd8667fb7d18562628b2d4e"}]},"branch":"refs/heads/master"},"47073baab62144bd69cc207ec9f9bf74bd512fbc":{"kind":"REWORK","_number":22,"created":"2022-06-09 20:00:00.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/22"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-06-09 19:59:35.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides Fibre Channel and iSCSI support for\nTATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nWe have tested it with several OpenStack versions, including Queens,\nVictoria, Train.\nWe have also tested it with Yoga and we are working to cover additional\nOpenStack releases.\n\nWe will be glad to join the OpenStack community and are looking for your\nfeedback on the driver.\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47073baab62144bd69cc207ec9f9bf74bd512fbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47073baab62144bd69cc207ec9f9bf74bd512fbc"}]},"branch":"refs/heads/master"},"910733638c523367c8f5c017b41077b712f582a4":{"kind":"REWORK","_number":23,"created":"2022-06-17 14:22:22.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/23"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-06-17 13:44:12.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides iSCSI support for TATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/910733638c523367c8f5c017b41077b712f582a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/910733638c523367c8f5c017b41077b712f582a4"}]},"branch":"refs/heads/master"},"ee73fbb7ffe64507d2afdf10ed9d00da8ef4df63":{"kind":"REWORK","_number":24,"created":"2022-07-11 06:43:47.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/24"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-07-06 13:13:50.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides iSCSI support for TATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nVolume Multiattachment\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\nHigh Availability\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee73fbb7ffe64507d2afdf10ed9d00da8ef4df63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee73fbb7ffe64507d2afdf10ed9d00da8ef4df63"}]},"branch":"refs/heads/master"},"0d2746015c2b2ccea0ea6cec2baf9afe1b9fd833":{"kind":"REWORK","_number":25,"created":"2022-07-12 06:30:41.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/25"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-07-12 06:30:01.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides iSCSI support for TATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nVolume Multiattachment\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\nHigh Availability\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0d2746015c2b2ccea0ea6cec2baf9afe1b9fd833"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0d2746015c2b2ccea0ea6cec2baf9afe1b9fd833"}]},"branch":"refs/heads/master"},"2a1c01c59eb58070fd6d0653589ee61b830769b6":{"kind":"REWORK","_number":26,"created":"2022-07-14 08:54:29.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/26"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-07-14 08:54:05.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides iSCSI support for TATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nVolume Multiattachment\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\nHigh Availability\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a1c01c59eb58070fd6d0653589ee61b830769b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a1c01c59eb58070fd6d0653589ee61b830769b6"}]},"branch":"refs/heads/master"},"7565b039717dd7d3c5263764c41eb5b36ccfe917":{"kind":"REWORK","_number":27,"created":"2022-07-21 16:49:18.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/27"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-07-21 16:48:47.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides iSCSI support for TATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nVolume Multiattachment\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\nHigh Availability\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7565b039717dd7d3c5263764c41eb5b36ccfe917"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7565b039717dd7d3c5263764c41eb5b36ccfe917"}]},"branch":"refs/heads/master"},"bc23474e8fd4e635ecade659d8f300e0e568d737":{"kind":"REWORK","_number":28,"created":"2022-07-25 11:54:58.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/28"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-07-25 10:41:35.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides iSCSI support for TATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nVolume Multiattachment\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\nHigh Availability\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc23474e8fd4e635ecade659d8f300e0e568d737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc23474e8fd4e635ecade659d8f300e0e568d737"}]},"branch":"refs/heads/master"},"3349c28d377f5ee930143a2dd3daefc1ab8bc9af":{"kind":"REWORK","_number":29,"created":"2022-07-25 16:50:08.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/29"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-07-25 16:49:55.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides iSCSI support for TATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nVolume Multiattachment\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\nHigh Availability\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3349c28d377f5ee930143a2dd3daefc1ab8bc9af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3349c28d377f5ee930143a2dd3daefc1ab8bc9af"}]},"branch":"refs/heads/master"},"ae5c2aeff7c88360e19e613fb21c4d0c4896152c":{"kind":"REWORK","_number":30,"created":"2022-07-26 12:57:25.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/30"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-07-26 12:56:17.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides iSCSI support for TATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nVolume Multiattachment\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\nHigh Availability\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae5c2aeff7c88360e19e613fb21c4d0c4896152c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae5c2aeff7c88360e19e613fb21c4d0c4896152c"}]},"branch":"refs/heads/master"},"02ff484fd2534511eedcf98cc838d7e379c34400":{"kind":"REWORK","_number":31,"created":"2022-07-26 17:28:31.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/31"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-07-26 17:28:21.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides iSCSI support for TATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nVolume Multiattachment\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\nHigh Availability\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/02ff484fd2534511eedcf98cc838d7e379c34400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/02ff484fd2534511eedcf98cc838d7e379c34400"}]},"branch":"refs/heads/master"},"6934463943f383c5c8d513720caf23f1cef83818":{"kind":"REWORK","_number":32,"created":"2022-07-27 07:06:03.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/92/825492/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/825492/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/825492/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/825492/32"}}},"commit":{"parents":[{"commit":"ee662725a61a12546dcff8f530a8b42762292192","subject":"Merge \"Add releasenotes to drop python3.6|7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ee662725a61a12546dcff8f530a8b42762292192"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-05-23 12:18:11.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-07-27 07:05:46.000000000","tz":180},"subject":"Initial commit for Yadro Tatlin.UNIFIED driver","message":"Initial commit for Yadro Tatlin.UNIFIED driver\n\nYADRO Cinder Driver\n\nThis commits introduces the YADRO Cinder Driver.\nYADRO Cinder Driver provides iSCSI support for TATLIN.UNIFIED storages.\nTATLIN.UNIFIED is a hybrid storage array with block and file access\nsupporting NVMe/SAS SSDs and SAS/NL-SAS/SATA HDDs designed to address\nmid-range needs of enterprise. It is actively developed and maintained\nby YADRO, a technology company with in-house R\u0026D focused on designing\nand developing efficient server and storage products for enterprise\nworkloads.\n\nThe YADRO Cinder Driver currently supports the following functionality:\n\nCreate Volume\nDelete Volume\nAttach Volume\nDetach Volume\nExtend Volume\nCreate Volume from Volume (clone)\nCreate Image from Volume\nVolume Migration (host assisted)\nVolume Multiattachment\nExtend an Attached Volume\nThin Provisioning\nManage/Unmanage Volume\nImage Cache\nHigh Availability\n\nImplements: blueprint yadro-tatlin-unified-driver\nCo-Authored-By: Vladislav Belogrudov \u003cv.belogrudov@yadro.com\u003e\nChange-Id: I586b79519573add80a1c34d9ba2656073852b4f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6934463943f383c5c8d513720caf23f1cef83818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6934463943f383c5c8d513720caf23f1cef83818"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[]}
