)]}'
{"id":"openstack%2Fcinder~826520","triplet_id":"openstack%2Fcinder~stable%2Fussuri~Iee14d5efd9286f478da373dcf2fa2c86aabc9975","project":"openstack/cinder","branch":"stable/ussuri","topic":"fix-sdk-func-job","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-01-27 16:14:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iee14d5efd9286f478da373dcf2fa2c86aabc9975","subject":"[stable-only] Specify openstacksdk stable branch","status":"MERGED","created":"2022-01-26 19:56:35.000000000","updated":"2022-01-28 10:28:52.000000000","submitted":"2022-01-27 16:14:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826520-fix-sdk-func-job","meta_rev_id":"b66cf169a10e0b9638ac71393dd8c7f7c06363b9","_number":826520,"virtual_id_number":826520,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-27 16:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-27 14:08:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2022-01-27 14:01:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-27 14:08:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-01-26 22:45:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-26 19:57:46.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-01-26 22:26:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-26 22:45:15.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-01-26 22:52:43.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-26 22:56:12.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-01-26 23:12:19.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-27 00:00:03.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-01-27 05:01:30.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-27 14:01:38.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-01-27 14:08:27.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2022-01-28 10:28:52.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"a4ee771e6c771cb4fa1a3a1c6097df4e1e118fe5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-01-26 19:56:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d21ffc04aa641d85e4b09599e052775ffbde0e22","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-01-26 19:57:46.000000000","message":"Patch Set 1:\n\nUnable to freeze job graph: Encryption/decryption failed.","accounts_in_message":[],"_revision_number":1},{"id":"6b0908b2203de1eb457344b8847cd19daffe872f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 22:26:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fa275b424b14355aa5efa467628d74b : SUCCESS in 8m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c95e895bdc7e4f05ab33785399d0139c : SUCCESS in 9m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9db3f47b06ca45f88e1049e7ccb8dfd3 : SUCCESS in 10m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/013b06f2ddb84eb3919c809c5d3c4cee : SUCCESS in 11m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/6baf6472179e4b3ca12179b32f83f92f : SUCCESS in 1h 11m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/da38e5bedbdd460b828952e95a1869b3 : SUCCESS in 1h 14m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5abfc3e8e47f41b5aa9f52edb16e54e5 : SUCCESS in 56m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cffe0a112fc346cb87dd72dcacdbf9dd : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/df5cad859ff54a32a8a6c64baa4cd9da : SUCCESS in 7m 27s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f31db92c74144a00a8b632732b33837d : SUCCESS in 6m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e859d6aa60ae4fa99326b3f9ad2bf06d : SUCCESS in 40m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/74e4f7f1ca32477daf8f2920d8674f01 : SUCCESS in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ddda8724abdf49dabfc863d90b44e53e : SUCCESS in 1h 23m 08s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/19c7c2a28499449e9b1f9c0ed44a2489 : SUCCESS in 1h 27m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cd5dabd71e6147faa1c54260b68a4744 : SUCCESS in 1h 25m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d80a3361fc144c3eabbc8a1474f046b7 : SUCCESS in 1h 23m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9b1586288e0744599e8e5651ef156a03 : FAILURE in 1h 21m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ea9e00217b5c4668bf31c48a6771e571 : SUCCESS in 2h 25m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7eb48a40d0a4fe4a8aa14f22434da11 : SUCCESS in 54m 46s","accounts_in_message":[],"_revision_number":1},{"id":"6de8fab506ddbcf7b1db535e2485f740b766026c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-01-26 22:45:04.000000000","message":"Patch Set 1: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af910e69d25f93cae4eb54f5f447b334476b4995","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-01-26 22:45:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/826520/1/upstream-solidfire-check/cinder-solidfire-iSCSI/608221a/ : FAILURE in 14m 20s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"829d79d2a4b72ab9260df27c8240a6fdf4887ec3","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-26 22:52:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/826520/1/upstream-check/cinder-cDOT-iSCSI/3b54410/ : FAILURE in 13m 17s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/826520/1/upstream-check/cinder-cDOT-NFS/28ed5b7/ : FAILURE in 8m 09s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/826520/1/upstream-check/cinder-cDOT-FCP/a9b17ef/ : FAILURE in 22m 55s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"9db9b4bbfa2a3ba7dc86adc0f7cec365fc06fdf7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-01-26 22:56:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/20/826520/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f53ca0e/ : FAILURE in 6m 39s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/20/826520/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/8ceb3e7/ : FAILURE in 7m 00s","accounts_in_message":[],"_revision_number":1},{"id":"75444a52a955b3a53e7da35a0d4e14913f5a43ac","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-26 23:12:19.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/20/826520/1/check/tempest-dsvm-full-bionic/df0f74f/ : SUCCESS in 2h 48m 31s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/20/826520/1/check/tempest-dsvm-full-bionic-py3/7fed13b/ : SUCCESS in 2h 29m 37s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/20/826520/1/check/devstack-plugin-ceph-dsvm-bionic-py3/aef81bf/ : FAILURE in 43m 57s","accounts_in_message":[],"_revision_number":1},{"id":"8ee4568316b8c0b2921638bb4ee6a9faac246b28","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-01-27 00:00:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/20/826520/1/check/iscsi-driver-alletra-6k/0105234/0105234.html : FAILURE in 50m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0a834436805fba293ee483f7c4b0ce660c008020","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-01-27 03:58:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.136/logs/20/826520/1/check/dsvm-tempest-cisco-zonemanager-job/29366d8/ : FAILURE in 28m 41s","accounts_in_message":[],"_revision_number":1},{"id":"586ac488573b7d8578a0b1471e2cf6d911dfcfe9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-27 05:01:30.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-20-826520-1/ns5_nfs : SUCCESS in 4h 25m 18s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-20-826520-1/ns5_iscsi : SUCCESS in 4h 23m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"8f5f3ef4b29d0993190757df000e0968b8d20cea","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-01-27 07:36:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/20/826520/1/check/iscsi-driver-nimble/a9b9246/a9b9246.html : FAILURE in 57m 45s","accounts_in_message":[],"_revision_number":1},{"id":"702b0b087fc686f0afa1d66cf69b91c334a56dc7","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-01-27 14:01:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1991828de0bdf0f097296ffe518f3a75b96c9719","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-01-27 14:08:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"96603f697606877ebaa3ce1b3e44d29d6ab60081","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 14:12:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aafcfdb690024c1ea83acacca765b835e7b9a08e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 16:14:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3f35110e3e44f1db2c1b9fd93a40e20 : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd9281e3497b4d61a9c76f136c1f4d8c : SUCCESS in 10m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6935d73b5064938b6d5607ec0333387 : SUCCESS in 13m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/097c223f42e542daad04028c2aa75cb1 : SUCCESS in 16m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/56d79bb1bab346968f53033e8d088c6c : SUCCESS in 1h 10m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6311caf4bb664876a21960f281739ee2 : SUCCESS in 1h 08m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18fa9d5b4020445ea79986ed998e8704 : SUCCESS in 53m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/40a409c4a41b4f4da9c9ffa4c27192bc : SUCCESS in 1h 17m 17s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7189b4892d1f42ae8cfd7053d8097d90 : SUCCESS in 1h 19m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eba78e2eeafa41d0857c1922985fd64c : SUCCESS in 51m 46s","accounts_in_message":[],"_revision_number":1},{"id":"38740db823abd0eb812798cb6533f00c570e6498","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 16:14:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"adece19a9b46f84572597b66a17cd2fc9f07e9a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 16:26:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a6d1b7a70af496eb0739ec506e277af : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b66cf169a10e0b9638ac71393dd8c7f7c06363b9","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-01-28 10:28:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/20/826520/1/ : RETRY_LIMIT in 2m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eeda43812c03face2fbe797766e09ff3b0f86136","revisions":{"eeda43812c03face2fbe797766e09ff3b0f86136":{"kind":"REWORK","_number":1,"created":"2022-01-26 19:56:35.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/20/826520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/826520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/826520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/826520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/826520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/826520/1"}}},"commit":{"parents":[{"commit":"37f61486c88206e351260a0b78c65f70d4fd3df9","subject":"[stable-only] Pin virtualenv and setuptools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/37f61486c88206e351260a0b78c65f70d4fd3df9"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-01-26 17:12:00.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-01-26 19:55:54.000000000","tz":-300},"subject":"[stable-only] Specify openstacksdk stable branch","message":"[stable-only] Specify openstacksdk stable branch\n\nTaking the advice given here:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2022-January/026905.html\nto fix the openstacksdk-functional-devstack job.\n\nNOTE: in stable/ussuri, this job was not specified in cinder\u0027s\n.zuul.yaml, but was brought in by the integrated-gate-storage template:\nhttps://opendev.org/openstack/tempest/src/commit/86db21ea6afb6c26a21fc143a7d061e947c31a93/zuul.d/integrated-gate.yaml#L382-L398\n\nSo we add the job here so that we can override its definition.\n\nChange-Id: Iee14d5efd9286f478da373dcf2fa2c86aabc9975\n(cherry picked from commit b66dbb151d367980f630e4688c4dfd8a465b671e)\n(cherry picked from commit f9d85e60748998dd9cd26f8be0690a99d4bca4c9)\n(cherry picked from commit ad392294a8ab538199c0670582aef34498cca51f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eeda43812c03face2fbe797766e09ff3b0f86136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eeda43812c03face2fbe797766e09ff3b0f86136"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
