)]}'
{"id":"openstack%2Fcinder~827860","triplet_id":"openstack%2Fcinder~master~Ide831a221a7561bbb7da193f0edb011f1989420e","project":"openstack/cinder","branch":"master","topic":"fix-request_id","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-02-09 13:16:29.000000000","reason":"Brian Rosmaita replied on the change"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2022-02-23 15:21:24.000000000","reason":"Eric Harney replied on the change"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-02-24 01:31:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ide831a221a7561bbb7da193f0edb011f1989420e","subject":"Fix request id mess on Cinder-API","status":"MERGED","created":"2022-02-04 13:08:36.000000000","updated":"2022-02-26 13:46:39.000000000","submitted":"2022-02-24 01:31:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827860-fix-request_id","meta_rev_id":"04efdf218f3da66c19931bffbe68df6d7040d297","_number":827860,"virtual_id_number":827860,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-24 01:31:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-02-23 20:06:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2022-02-23 17:04:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-02-23 20:06:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-04 13:40:32.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-04 13:40:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-04 13:46:33.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-04 14:21:44.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2022-02-04 14:34:37.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-04 14:48:22.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-02-04 14:48:50.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-02-04 14:49:59.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-02-04 14:57:12.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-04 15:35:33.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-02-04 15:36:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-04 15:44:16.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-02-04 15:50:05.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-02-04 15:53:56.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-04 16:31:48.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-02-04 16:33:02.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-04 16:34:06.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-04 17:31:23.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-02-04 17:33:37.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-04 17:34:09.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-04 17:41:02.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-02-04 18:06:01.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-02-04 18:09:20.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-02-04 18:16:19.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-02-04 21:45:34.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-02-08 12:40:44.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-08 19:48:04.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-02-08 20:15:59.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-02-08 20:55:13.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-22 19:06:16.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"CC"},{"updated":"2022-02-23 23:42:25.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"}],"messages":[{"id":"8eb7e7e658c77a94b2dcc2b281a2c88568742bac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-02-04 13:08:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f8ddc7149e5224c9397f29539736742c2edeb10","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-04 13:40:32.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/60/827860/1/check/DellEMC_XtremIO_FC/1a245e8/DellEMC_XtremIO_FC/648 : FAILURE in 14m 25s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/60/827860/1/check/DellEMC_XtremIO_iSCSI/ccfbbb2/DellEMC_XtremIO_iSCSI/658 : FAILURE in 14m 40s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"d04df4aed728db021bb78d91179749cbe8d220dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 13:40:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/495e8466e6c54cfdb75933c5184f0b1f : FAILURE in 25m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e13c451c966b40a89745f8a0671fdb36 : FAILURE in 25m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0c8c71df52ee522832b3589882788b147ddbbc4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-04 13:46:33.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15141/","accounts_in_message":[],"_revision_number":1},{"id":"a1d873adf2d855652b54ef610b82f2000280202a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-04 13:47:32.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15141/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"805b37b8797b19f3e5184f161989345dd6687428","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-02-04 14:34:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/827860/1/check/iscsi-driver-primera/28b7ec1/28b7ec1.html : SUCCESS in 1h 18m 11s","accounts_in_message":[],"_revision_number":1},{"id":"cefd4fb60a05368603680ae23aee5eb1a20e0b87","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-04 14:48:22.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/827860/1/iscsi/ : FAILURE in 1h 39m 13s\n- cinder-smb http://cloudbase-ci.com/cinder/827860/1/smb/ : SUCCESS in 1h 09m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7946b4546b7aa15d2778a0c6c706295b0a7dd9fa","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-02-04 14:48:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/827860/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d89b0b7/ : SUCCESS in 1h 38m 37s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/827860/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/2f65107/ : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":1},{"id":"2b023feb2a0ff460c3e4a7bcf66179a6d6bc47c9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-04 14:49:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/60/827860/1/check/iscsi-driver-alletra-6k/df80a12/df80a12.html : SUCCESS in 1h 39m 59s","accounts_in_message":[],"_revision_number":1},{"id":"576050f7421c84f23700b543dd732e72d8b9fb1e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-04 14:57:12.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/827860/1/check/tempest-dsvm-full-focal-py3/da94184/ : FAILURE in 22m 48s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/827860/1/check/devstack-plugin-ceph-dsvm-focal-py3/84aeca4/ : FAILURE in 23m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b7edbd2a18b8e73c910f8de73104414bb8ba038c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-02-04 15:35:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/827860/1/20220204221614/NEC-V-FC : SUCCESS in 1h 56m 13s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/827860/1/20220204221614/NEC-V-ISCSI : SUCCESS in 2h 19m 15s","accounts_in_message":[],"_revision_number":1},{"id":"22bf76f7bb29985f1cd0447b2991cfce01baa13b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 15:36:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f31604fa5dc4bcc9b1091c973402d28 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/516f0e57c21e405982eb66a7f2ea6aac : SUCCESS in 10m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5661e5cbdb33487eb1c724391393b721 : SUCCESS in 11m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a8e9f7334f545e6af6dec08ff0fa442 : SUCCESS in 14m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/2bc74324037a429290f2d520ecd6b3f9 : SUCCESS in 56m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/47e1663bd95441d193a8d875abdf5a89 : SUCCESS in 1h 27m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98855717d7d64d22abc663dffae5c5c8 : SUCCESS in 48m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1b56a93187e470d8e163daf08784a25 : SUCCESS in 7m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f42e734713e34cfe989852847dce22f3 : SUCCESS in 15m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3bf116e7fd1a43dc8d574d65c0a5d7d9 : SUCCESS in 6m 44s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e87687621b31421c85c4f4dd1a9cb889 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/037a5310c47a476e92361941b5761a39 : SUCCESS in 5m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/660fb70408d045dbabeba7e77ace0017 : SUCCESS in 7m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8828041b66aa40e194a047ef25d2e68a : SUCCESS in 36m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6f37f37e49b9489f8debf7a73e8f59ef : SUCCESS in 7m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9e1cc500dc6a46e595216f8b7a55ffc1 : SUCCESS in 43m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2a6ffdfb7bbf4719b62b72038983ba23 : SUCCESS in 1h 29m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a70a5c4f474a46488245f89c1a1dadb7 : SUCCESS in 1h 30m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f0c935851f224fb387c3132d6444e7e7 : SUCCESS in 1h 03m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/309a336885cd4586b367409e72ff0f61 : SUCCESS in 1h 16m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/46e23956e16a4dcd878fc05bf3c07173 : SUCCESS in 34m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d22bc190f4a54ff49c53fbe4a718ebc0 : SUCCESS in 1h 13m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c8f7f4d7e81849a293ca70b09d4ef296 : SUCCESS in 2h 08m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3130d8325e674a9a8899ed750db8b52a : SUCCESS in 45m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3d4d45f4f86c2480b202121912a701f44dfd7bb2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-02-04 15:44:16.000000000","message":"Patch Set 1:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/04_02_2022_15_36_20\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"76649737a271677822aaaf573c8772d6cf5f8865","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-02-04 15:50:05.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/60/827860/1/check/DellEMC_PowerStore_iSCSI/badc38c/DellEMC_PowerStore_iSCSI/316 : FAILURE in 12m 59s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/60/827860/1/check/DellEMC_PowerStore_FC/12361a8/DellEMC_PowerStore_FC/333 : FAILURE in 13m 31s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"193a7239eaaba5998276ea04e58315a94a014cf9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-04 15:53:56.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/60/827860/1/check/DellEMC_PowerFlex/2f22129/DellEMC_PowerFlex/366 : FAILURE in 11m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"f3c6699296fec4209af48820ae943305d130e2c4","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-04 15:56:40.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/60/827860/1/check/DellEMC_XtremIO_FC/0907ae1/DellEMC_XtremIO_FC/649 : FAILURE in 15m 03s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/60/827860/1/check/DellEMC_XtremIO_iSCSI/993af9d/DellEMC_XtremIO_iSCSI/659 : FAILURE in 14m 50s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"845911e94296c91406e740e00b4a5e2474ab6428","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-02-04 16:31:48.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/827860/1/upstream-solidfire-check/cinder-solidfire-iSCSI/7d0aedb/ : FAILURE in 45m 32s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"6f26c57fa5d401496b7a8b6f092e9c4981544e0e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-04 16:33:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/827860/1/upstream-check/cinder-cDOT-iSCSI/00f4de1/ : SUCCESS in 32m 05s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/827860/1/upstream-check/cinder-cDOT-NFS/b8644c4/ : FAILURE in 53m 50s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/827860/1/upstream-check/cinder-cDOT-FCP/cc08ee7/ : SUCCESS in 54m 47s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c2a502f9dfb8d7babeaa9a63d4992b879fefa2b5","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-02-04 16:34:06.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/60/827860/1/check/DellEMC_PowerMAX_FC/df2d343/DellEMC_PowerMAX_FC/278 : FAILURE in 52m 18s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/60/827860/1/check/DellEMC_PowerMAX_iSCSI/ae6f2aa/DellEMC_PowerMAX_iSCSI/291 : FAILURE in 27m 09s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"e95835abaa98350d5a17d31a49df3a0e0a2fa0fa","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-02-04 17:31:23.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/60/827860/1/check/DellEMC_SC_FC/4e4a0fc/DellEMC_SC_FC/300 : FAILURE in 1h 05m 55s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/60/827860/1/check/DellEMC_SC_iSCSI/c9d0fac/DellEMC_SC_iSCSI/296 : SUCCESS in 1h 54m 53s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"e84c405a48ca25b8e20c11a7f6a0f8f2f400ea67","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-04 17:33:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/827860/1/ : SUCCESS in 1h 47m 17s","accounts_in_message":[],"_revision_number":1},{"id":"cc092f64da6c8b76d3b0aa5cd09dce65a18a910c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-04 17:34:09.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-827860-1/ns5_nfs : SUCCESS in 4h 19m 46s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-827860-1/ns5_iscsi : SUCCESS in 4h 24m 29s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"414f730999bb1ea6210ab37842b2dc34061b1f74","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-04 17:41:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/827860-1/HPMSA-iSCSI/ : SUCCESS in 51:14.77\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/827860-1/HPMSA-FC/ : SUCCESS in 43:01.46","accounts_in_message":[],"_revision_number":1},{"id":"6fb5ef510970fe6b1a7315c7def221f0c96b6dc7","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-04 18:06:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/827860-1/Lenovo-iSCSI/ : SUCCESS in 1:01:52\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/827860-1/Lenovo-FC/ : SUCCESS in 56:53.20","accounts_in_message":[],"_revision_number":1},{"id":"5f82e150476a98328f70a3f38338a77186761603","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-04 18:09:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/827860-1/PVME-iSCSI/ : SUCCESS in 1:04:23\n- PVME-FC https://os-logs.seagate.com/dell-ci/827860-1/PVME-FC/ : SUCCESS in 1:00:18","accounts_in_message":[],"_revision_number":1},{"id":"e979ebda4f5cb4da3ace8d0eae2a68884c43aa3a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-04 18:13:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/827860/1/check/iscsi-driver-nimble/691a6b3/691a6b3.html : SUCCESS in 1h 55m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3d2e0dffcb483d730415b3b45c2627832b2d891a","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-02-04 18:16:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/827860-1/Seagate-iSCSI/ : SUCCESS in 1:05:45\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/827860-1/Seagate-FC/ : SUCCESS in 1:03:39","accounts_in_message":[],"_revision_number":1},{"id":"3d8cd5d6e8d7bd1f6dc3ae0aa580d39826828807","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-04 21:45:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/60/827860/1/check/dsvm-tempest-cisco-zonemanager-job/ddb2335/ : FAILURE in 1h 08m 31s","accounts_in_message":[],"_revision_number":1},{"id":"cc70f05ffa0d76bbf526dcbf47c5bbe11aaff5c3","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-02-08 11:44:53.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e2d9ea3c19319f3fc304a924b46de13a3a51828","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-02-08 12:40:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/827860/1/20220208181209/NEC-FC : SUCCESS in 3h 26m 28s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/827860/1/20220208181209/NEC-ISCSI : SUCCESS in 3h 28m 31s","accounts_in_message":[],"_revision_number":1},{"id":"81c20630e4e8ca1ffbe39cfc6366d5279a4c7239","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-02-08 15:47:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a924357e2b915413e8ae008851a4a6c982758c35","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-08 15:48:24.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15157/","accounts_in_message":[],"_revision_number":2},{"id":"0719dc3e908e8a344e137482dcc91de1970700bd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-08 15:49:53.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15157/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"272d6da0ce6492423c2a9f69f58884f1bbbbb060","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-08 16:29:37.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/60/827860/2/check/DellEMC_XtremIO_FC/80a3ae8/DellEMC_XtremIO_FC/678 : FAILURE in 18m 23s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/60/827860/2/check/DellEMC_XtremIO_iSCSI/6240e8c/DellEMC_XtremIO_iSCSI/688 : FAILURE in 19m 04s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"a0349d6f52b03f4cd403271b13e3f805cc975c50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 16:41:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ecfc093d7da413b8dac8a6600252ff0 : SUCCESS in 20m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f35927c4225e45f3be51d3a0c423bf12 : FAILURE in 25m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"02acad559f03d9887280c0722328b1c9f76b8fed","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-08 16:58:52.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/827860/2/iscsi/ : SUCCESS in 1h 11m 17s\n- cinder-smb http://cloudbase-ci.com/cinder/827860/2/smb/ : SUCCESS in 1h 07m 04s","accounts_in_message":[],"_revision_number":2},{"id":"db37c9fbbcb4b417240244460d26994473b48aee","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-08 17:06:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/60/827860/2/check/dsvm-tempest-cisco-zonemanager-job/7b662b7/ : SUCCESS in 1h 04m 11s","accounts_in_message":[],"_revision_number":2},{"id":"5ee55dd45616efbbd4b84a8c14ccf2ee3c724976","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-02-08 17:19:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/60/827860/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a5e276d/a5e276d.html : FAILURE in 1h 18m 25s","accounts_in_message":[],"_revision_number":2},{"id":"e3c9eaa1959dad0216e78fd1b70bc1d1dba880a9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-02-08 17:20:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/827860/2/check/iscsi-driver-primera/a65611c/a65611c.html : SUCCESS in 1h 20m 33s","accounts_in_message":[],"_revision_number":2},{"id":"018a37a8f036c24c568212fde4c3a3c8efb5cc12","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-08 17:26:12.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/60/827860/2/check/iscsi-driver-alletra-6k/9f4b3f8/9f4b3f8.html : SUCCESS in 1h 37m 55s","accounts_in_message":[],"_revision_number":2},{"id":"133af3c440018bfd261f7ed5daa5b68e6ce60162","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-08 17:41:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/827860/2/check/iscsi-driver-nimble/3623cdb/3623cdb.html : SUCCESS in 1h 53m 28s","accounts_in_message":[],"_revision_number":2},{"id":"e1443349dbc36e07b62032580a02f95852b02465","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-02-08 18:24:16.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/827860/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/75bb60b/ : FAILURE in 2h 35m 25s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/827860/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/fba2d1b/ : SUCCESS in 57m 49s","accounts_in_message":[],"_revision_number":2},{"id":"9d58e42c8afdc376f5cc7d2241bc3305da73761b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-08 18:25:27.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/827860/2/ : SUCCESS in 2h 20m 58s","accounts_in_message":[],"_revision_number":2},{"id":"1f9a654eda602d97659bbcf2edd5cde12669d9b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 18:58:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e69dd9c47449435093f17543e98c730f : SUCCESS in 9m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/809e0dbe185e4e628e7ba7f8ef6eed62 : SUCCESS in 10m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73e5f5935d7943c889985cb765b28579 : SUCCESS in 12m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34e4ea36104b4245b750f00ca5c49538 : SUCCESS in 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/9007d9f6f03a4016bc9ca44e67a6936c : SUCCESS in 55m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c2e9100c881b4047b05c5776fc74fa11 : SUCCESS in 1h 17m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d25ee6718b4e40879a51d5d496fe0eb3 : SUCCESS in 48m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0a2e9f4990c4db4af86220d59afed8a : SUCCESS in 7m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e5bb189a1f8e4664b458c17828e99346 : SUCCESS in 13m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/28303de4b4514bd9a5673f94175008aa : SUCCESS in 8m 00s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e97fd0bde7394f9b8afe5c046953980b : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/68fa1aaab9044a3c9de77885bfa02d81 : SUCCESS in 6m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/870bfaaf3c204f33bad4861e3cd435dc : SUCCESS in 8m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eb62be6b9b5444feaabf44ad9b500239 : SUCCESS in 37m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0f79ebd33c2442d6bb0f47746c5ab87c : SUCCESS in 7m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/63dddd84395747a492e9bd722380794e : SUCCESS in 1h 17m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1a0131d530084fb8b138e49dc8322386 : SUCCESS in 1h 29m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5568615ae7a44f4fbc0514d4027b2981 : SUCCESS in 1h 30m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/09da765fdd264b3f9cab58a663f99a42 : SUCCESS in 1h 04m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b6011bf6833b4047abc88bba2bc83970 : SUCCESS in 1h 21m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a82c6350af37436fb3693ddb7b3263a7 : SUCCESS in 34m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0a5278b7224a4150813675e2669a54a1 : SUCCESS in 1h 26m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/77f38324646d4df091c05af81e01839e : SUCCESS in 2h 25m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53513ffa35584b7ab7f1f14c6c1045b4 : SUCCESS in 53m 39s","accounts_in_message":[],"_revision_number":2},{"id":"a8c7e7f2bb779d5ba668a5a585338c7fd399e745","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-02-08 19:07:06.000000000","message":"Patch Set 2:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/08_02_2022_18_58_47\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"65581f23d76c319c8f113f92a2202d44502edecb","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-08 19:12:09.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"c633ee8688e48f523d559e4a0371c1381a0a56b3","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-08 19:12:09.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"3516c80c9dbc69acb0e9ee8ef7632d8a41f82e8e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-02-08 19:14:44.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/60/827860/2/check/DellEMC_PowerStore_iSCSI/893cc8a/DellEMC_PowerStore_iSCSI/329 : FAILURE in 13m 23s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/60/827860/2/check/DellEMC_PowerStore_FC/bc26b8d/DellEMC_PowerStore_FC/346 : FAILURE in 14m 06s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"9d9b768b27e23b2e8d2869ecf9e7555cfa9cae88","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-08 19:35:50.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/827860/2/check/tempest-dsvm-full-focal-py3/fb07775/ : SUCCESS in 1h 56m 13s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/827860/2/check/devstack-plugin-ceph-dsvm-focal-py3/6b028e1/ : SUCCESS in 2h 51m 43s","accounts_in_message":[],"_revision_number":2},{"id":"32db15f8415f2ceec9054a737741bf58f90eaebe","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-08 19:48:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-60-827860-2/1644346718 : SUCCESS in 34m 05s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"074ecfe2a158d9fc604fccbed43ecfce0cb0b6fe","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-08 20:15:59.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e1c5db248662b7dc58de2ee2bcd4e195fb5a99d5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-08 20:51:36.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-827860-2/ns5_nfs : SUCCESS in 4h 28m 41s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-827860-2/ns5_iscsi : SUCCESS in 4h 25m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"60263b32361b0cd429337f6cd21d799c732086c4","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-02-08 20:55:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/60/827860/2/check/dsvm-tempest-zadara-driver/8c70655 : SUCCESS in 1h 05m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"384915d5418be4ac9caa55d5f58a9380b3518a5d","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-08 21:32:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/827860-2/PVME-iSCSI/ : SUCCESS in 1:00:48\n- PVME-FC https://os-logs.seagate.com/dell-ci/827860-2/PVME-FC/ : SUCCESS in 58:10.25","accounts_in_message":[],"_revision_number":2},{"id":"6ce138dede6764b186ca5631c2afae302980fd05","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-08 21:37:25.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/60/827860/2/check/DellEMC_XtremIO_FC/7007649/DellEMC_XtremIO_FC/681 : FAILURE in 15m 24s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/60/827860/2/check/DellEMC_XtremIO_iSCSI/a1d564a/DellEMC_XtremIO_iSCSI/691 : FAILURE in 15m 36s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"54c923d4f53c3278eddacaf8f311726add53e2ed","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-08 21:40:32.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/60/827860/2/check/DellEMC_PowerFlex/9c87d15/DellEMC_PowerFlex/379 : FAILURE in 12m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"85cddabe60b1e21f5da69b9bd4c4d733af8e4257","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-02-08 21:43:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/827860/2/20220209004707/NEC-V-FC : SUCCESS in 1h 58m 27s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/827860/2/20220209004707/NEC-V-ISCSI : SUCCESS in 5h 56m 36s","accounts_in_message":[],"_revision_number":2},{"id":"b330ba9ade6a63db437b31e279a321447edf33ea","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-02-08 21:50:22.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/60/827860/2/check/DellEMC_PowerMAX_FC/5584963/DellEMC_PowerMAX_FC/291 : FAILURE in 52m 34s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/60/827860/2/check/DellEMC_PowerMAX_iSCSI/ae0c5f9/DellEMC_PowerMAX_iSCSI/304 : FAILURE in 27m 34s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"be81d18bd8d9b33d3c649338d5dbbcac8d17c1af","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-02-08 22:17:16.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/60/827860/2/check/DellEMC_SC_FC/1b20ad8/DellEMC_SC_FC/313 : FAILURE in 1h 12m 10s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/60/827860/2/check/DellEMC_SC_iSCSI/48430cc/DellEMC_SC_iSCSI/309 : SUCCESS in 1h 46m 51s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"ce2efbed63d5fa9625cf31fd43acb8c275a2719a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-09 03:20:56.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-827860-2/ns5_nfs : SUCCESS in 4h 40m 25s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-827860-2/ns5_iscsi : SUCCESS in 4h 40m 21s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"04c90d960d51544d0e3e5f77922d704f7b5b9c8c","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-02-09 12:14:53.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"77841cd4603809080655e315830a70583c0d048d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-02-09 12:15:00.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"13ff23e57017643602fa1e5eab1931680aff8524","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-09 12:30:28.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/60/827860/3/check/DellEMC_XtremIO_FC/63070bd/DellEMC_XtremIO_FC/699 : FAILURE in 15m 03s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/60/827860/3/check/DellEMC_XtremIO_iSCSI/aaac855/DellEMC_XtremIO_iSCSI/709 : FAILURE in 14m 55s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"0768a0f01da3aba7db416ceba9c96d22d5431490","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-09 12:51:11.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/827860/827860/3/check-not-vote/ext-cinder-zuul/4aa39d5 : FAILURE in 6m 12s","accounts_in_message":[],"_revision_number":3},{"id":"1233161b7404dece285ec56999cb426ee8d15a1e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-09 13:16:29.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"87ef45295fe1e1e2a207cdf820cae279bc638911","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 13:16:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2c64c6e75284bb28a6b07d1285ee274 : SUCCESS in 19m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d18bbba9137a4509ae152187146a1610 : SUCCESS in 23m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"88d35adee570ff99be98821417bb05a62c133298","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-09 13:25:33.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/60/827860/3/check/dsvm-tempest-cisco-zonemanager-job/c837c9b/ : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":3},{"id":"23cd8093fc2f7740b5dd65fad52f6498852bbb81","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-09 13:25:58.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/827860/3/iscsi/ : SUCCESS in 1h 10m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/827860/3/smb/ : SUCCESS in 48m 45s","accounts_in_message":[],"_revision_number":3},{"id":"42e0c30f5b2d39c5d9fe67740d170c936ea1cf5f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-02-09 13:43:19.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/827860/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b34d244/ : SUCCESS in 1h 26m 34s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/827860/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/574a7a0/ : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":3},{"id":"a02ad4e5834990834fc208e5ff0741bcbfb5b5aa","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-09 13:55:40.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/60/827860/3/check/iscsi-driver-alletra-6k/c8a8e6c/c8a8e6c.html : SUCCESS in 1h 39m 27s","accounts_in_message":[],"_revision_number":3},{"id":"7f61e5210b6707ff79f438ba2a11eb02a93556c5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-02-09 14:01:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/827860/3/check/iscsi-driver-primera/bfbf594/bfbf594.html : FAILURE in 1h 41m 14s","accounts_in_message":[],"_revision_number":3},{"id":"ae695e1db2fc77628643a26715d0b695d8eacb1a","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-02-09 14:06:53.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/60/827860/3/check/dsvm-tempest-zadara-driver/f39e557 : SUCCESS in 1h 04m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e75bfaa09b8614cc6e1fb3f5a132bc8f6d047536","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 14:40:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0b0ae86c7e049e3ae2b577db50c6b01 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5877c904538d417a901bb6126b4cfa5b : SUCCESS in 11m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77c3c2784ce74da8970dc83b792461a3 : SUCCESS in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2182dae25fb344db9a25e7e9cd10da86 : SUCCESS in 10m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/fa4518804a35489b841b73270ee428bd : SUCCESS in 58m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6cb39f68b8a348a2b0ff0d5f52166dd2 : SUCCESS in 1h 31m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/809eef3242074d3491516aaa1c9709e8 : SUCCESS in 47m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1e5586946564cefbd139389a2025e60 : SUCCESS in 6m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ff8b10a43f034eeaaf9d09e3c426a1fe : FAILURE in 12m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9c09248551c643a9ac76a82f8ce5bccc : SUCCESS in 5m 34s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/10f0e6d66f364b518a53a7137861b88e : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c3f33da926b74c15a1bf1c511f6d8e4f : SUCCESS in 6m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6642830aae034ab19e44277027123443 : SUCCESS in 8m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5be50999a0dc4c0a9d4330a9cf3ff6be : SUCCESS in 37m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/146aa808f54e45c688094e0b1501a90e : SUCCESS in 5m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3b88b136bb4441679e9bd12a9180e1ed : SUCCESS in 1h 09m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/27b11eab5c9b404aa55114d130c3b720 : SUCCESS in 1h 20m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bb79b85f39b34edeb350c5be2a6967a1 : SUCCESS in 1h 52m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e126a596e21540158c5a245d44a6725b : SUCCESS in 1h 01m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6446e66cc8c646ab89f0b50154121b27 : SUCCESS in 1h 17m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/491b6aafb52f4b85a1197a977b3d02c4 : SUCCESS in 43m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/34dc26b8b67e4468889ee78d1f851075 : SUCCESS in 1h 14m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7318b62586bb44318274e5d88e695a5c : SUCCESS in 2h 11m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a14054c21c1046648669ce8479ba28d4 : SUCCESS in 48m 29s","accounts_in_message":[],"_revision_number":3},{"id":"5fcf42b078810f7f2bde2cff5cffef1bb0c3db80","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-02-09 14:48:57.000000000","message":"Patch Set 3:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/09_02_2022_14_40_49\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"25b8bc91f6177a7ddd7b201286db6ba2a0d08f76","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-02-09 15:04:38.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/60/827860/3/check/DellEMC_PowerStore_iSCSI/eaa5506/DellEMC_PowerStore_iSCSI/339 : FAILURE in 13m 24s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/60/827860/3/check/DellEMC_PowerStore_FC/13f6f1a/DellEMC_PowerStore_FC/356 : FAILURE in 13m 56s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"c860857e8ae192ea205937c476b778c38a80b43f","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-09 15:11:23.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-60-827860-3/1644417638 : SUCCESS in 16m 55s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":3},{"id":"e52f41a515c8caa96e6fad8862ec20cc7e64f1ce","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-09 15:29:04.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/60/827860/3/check/DellEMC_PowerFlex/0cf61a1/DellEMC_PowerFlex/390 : FAILURE in 11m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"0d1c864cf9955d2795cd133612028705653b76d9","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-09 15:31:58.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/60/827860/3/check/DellEMC_XtremIO_FC/68ee403/DellEMC_XtremIO_FC/701 : FAILURE in 14m 50s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/60/827860/3/check/DellEMC_XtremIO_iSCSI/5c285f4/DellEMC_XtremIO_iSCSI/711 : FAILURE in 14m 35s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"db55bc70f2860fb876383d4cfdef8901704c910e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-09 15:44:19.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/60/827860/3/check/iscsi-driver-nimble/f1f726c/f1f726c.html : SUCCESS in 1h 49m 13s","accounts_in_message":[],"_revision_number":3},{"id":"051750d0f3fe8ca94af69dd7e52774708061201d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-09 15:45:38.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/827860/3/check/tempest-dsvm-full-focal-py3/1648c04/ : SUCCESS in 2h 00m 50s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/827860/3/check/devstack-plugin-ceph-dsvm-focal-py3/71308b9/ : SUCCESS in 3h 06m 18s","accounts_in_message":[],"_revision_number":3},{"id":"051e3a288790df3cd1fb033b6c53e4074665edf2","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-02-09 16:30:50.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/60/827860/3/check/DellEMC_PowerMAX_FC/d0a2efc/DellEMC_PowerMAX_FC/301 : FAILURE in 52m 08s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/60/827860/3/check/DellEMC_PowerMAX_iSCSI/66e0daa/DellEMC_PowerMAX_iSCSI/314 : FAILURE in 26m 28s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"46ce9b54faa01b5890975bf1234abbc1113e9192","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-09 17:31:35.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-827860-3/ns5_nfs : SUCCESS in 4h 41m 17s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-827860-3/ns5_iscsi : SUCCESS in 4h 49m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"12406b2442becc7c7f8c8efc40c42f8d173cbcc9","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-09 18:52:08.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/827860-3/HPMSA-iSCSI/ : SUCCESS in 52:25.75\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/827860-3/HPMSA-FC/ : SUCCESS in 43:38.29","accounts_in_message":[],"_revision_number":3},{"id":"605a84089626e3d7912ea8d6d15cda0f337c8dde","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-02-09 19:57:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/827860-3/Seagate-iSCSI/ : SUCCESS in 1:05:07\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/827860-3/Seagate-FC/ : SUCCESS in 1:03:38","accounts_in_message":[],"_revision_number":3},{"id":"058ee5376353e90a917fbd9105aacefa7f1add2a","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-02-09 20:13:23.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/60/827860/3/check/DellEMC_SC_FC/d52983a/DellEMC_SC_FC/323 : FAILURE in 1h 05m 25s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/60/827860/3/check/DellEMC_SC_iSCSI/7a245d5/DellEMC_SC_iSCSI/319 : SUCCESS in 2h 01m 54s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"8e61f25ad65e00e3b214a51829d7d8fa53369ecb","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-09 20:30:19.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/827860-3/Lenovo-iSCSI/ : SUCCESS in 59:48.22\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/827860-3/Lenovo-FC/ : SUCCESS in 56:03.42","accounts_in_message":[],"_revision_number":3},{"id":"ca36be50b4719dbe723717e24774e33fcfa906e7","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-09 22:33:00.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/827860/3/ : SUCCESS in 2h 06m 56s","accounts_in_message":[],"_revision_number":3},{"id":"d9bf728b0cefed653313982d9f2731330c78b472","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-09 22:58:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/827860-3/PVME-iSCSI/ : SUCCESS in 1:03:21\n- PVME-FC https://os-logs.seagate.com/dell-ci/827860-3/PVME-FC/ : SUCCESS in 1:00:49","accounts_in_message":[],"_revision_number":3},{"id":"46612056d54cce08b13444401b4d3fbf593e3700","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-02-09 23:43:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/827860/3/20220210063034/NEC-V-FC : SUCCESS in 1h 56m 52s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/827860/3/20220210063034/NEC-V-ISCSI : SUCCESS in 2h 12m 58s","accounts_in_message":[],"_revision_number":3},{"id":"25365990da4cc17f0d2ff03a639fad027af5035d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-02-11 10:06:23.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/60/827860/3/20220211154756/NEC-ISCSI : SUCCESS in 3h 15m 20s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/60/827860/3/20220211154756/NEC-FC : SUCCESS in 3h 18m 24s","accounts_in_message":[],"_revision_number":3},{"id":"a03607d237c260bbeff6791f5c224ca3da7fbf17","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-02-22 14:49:42.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5d90b4ffd4b01e32796178195f6728de4bb13b45","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-02-23 14:58:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"81a59489a174ba828d007c40b3e00e1b7a6904e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-02-23 15:13:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9f3f649530ece94ae9bc6c59d2735b47d8464aac","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-02-23 15:13:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c74501b604c933c7d27c5d494e58456d8d0b99ab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-23 15:13:26.000000000","message":"Patch Set 4:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15297/","accounts_in_message":[],"_revision_number":4},{"id":"39cff38ac019debcc50b7e5e8c5bcc1270a78c0b","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-02-23 15:13:46.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/60/827860/4/check/dsvm-tempest-zadara-driver/a62fce6 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 9s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1bae97e052733aa3fc66848ccf1982351f1d5191","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-23 15:15:05.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15297/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"ab2ab7d66b6e56d3a3fe66852a7c8e4514a3b5a4","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-02-23 15:21:24.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2b5c321f9fe6ecc8ab7462676ef525a7e5815c64","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-23 15:27:54.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/827860/4/iscsi/ : FAILURE in 13m 57s\n- cinder-smb http://cloudbase-ci.com/cinder/827860/4/smb/ : FAILURE in 13m 56s","accounts_in_message":[],"_revision_number":4},{"id":"6721be93166b55651325f435f6c0498b07214135","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-02-23 15:49:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1ac953d4886217e37400300e45b536e1b9f727c2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-23 16:04:15.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/827860/5/iscsi/ : FAILURE in 13m 55s\n- cinder-smb http://cloudbase-ci.com/cinder/827860/5/smb/ : FAILURE in 14m 04s","accounts_in_message":[],"_revision_number":5},{"id":"51478f3565b187f59535017fe9ef8c13e1e1e328","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 16:27:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0025eb1bb404e4e8c10fcf4a482aa30 : FAILURE in 30m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb34d175cb1a4eb2875a6c117a8a3239 : FAILURE in 31m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"db9bf528d6a8d530cd31d4eda7a88d0099149164","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-02-23 17:04:09.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f578f5f5f54e50e2ed7c5d5de5b77ac8d3015128","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-23 17:54:48.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/827860/827860/5/check-not-vote/ext-cinder-zuul/bff0802 : FAILURE in 13m 26s","accounts_in_message":[],"_revision_number":5},{"id":"bb80c36573674807d6238890ad4d3b2cbb258a94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 18:11:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b3cb1d27000413cb0d860843cf214f8 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9915ef2de0a4af39de45e9af4e1f7e5 : SUCCESS in 11m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/514c0d69cd394109b4b3488d69d27f07 : SUCCESS in 12m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaa5a5f244ff489e94ff1b0e10aa6eb7 : SUCCESS in 15m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/70d5e344733e4cd5ae88f6e66467d579 : SUCCESS in 59m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/756be37c6d3d4f12884204a1e2e6f738 : SUCCESS in 1h 13m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/359abab33b874ea5944294e9527ce9a3 : SUCCESS in 1h 04m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e9529ecaabc4fd4aea8842a32fd09d0 : SUCCESS in 7m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/65ade94afca14d3ba132f9d0862eb6a3 : SUCCESS in 15m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a2e22ec5380545b39f9144a49497c190 : SUCCESS in 6m 22s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eda7bbbbacff4dd1923d74726bb931c2 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b3df23ff9794484ead4dcd5be8238511 : SUCCESS in 6m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4880dff6173b4a3ba288191e1c1a75af : SUCCESS in 7m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/944338747bab4109ba21ef0b20e1b0eb : SUCCESS in 38m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a45d72626cdf4addb699d0dc323176ab : SUCCESS in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/68d3ecb8f1264ddab19ebb59d8973407 : SUCCESS in 1h 10m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/27015a8d3340451d96eb8a2c2cc3b2fa : SUCCESS in 1h 21m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7a69203316174aa6b5da4b13058e3c9c : SUCCESS in 1h 31m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ef47f42e64a1439c93471fb5772f5045 : SUCCESS in 1h 06m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/16b20b22de1a42a384fc903c2ea5a38a : SUCCESS in 1h 07m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/00283c6ca9f24c7c8c540b3d16b0d009 : SUCCESS in 40m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8679f9bfde7f42fdba13d0a495444aa4 : SUCCESS in 1h 09m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8623d85ef986462bb0a1a2b8170aea63 : SUCCESS in 2h 18m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11df8e14f5a844f18b31fde3ee2fc7f1 : SUCCESS in 47m 41s","accounts_in_message":[],"_revision_number":5},{"id":"808d305e23584e8e53bc6622dc5748802d7294e8","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-02-23 18:25:44.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/60/827860/5/check/dsvm-tempest-zadara-driver/7615cfd : SUCCESS in 1h 07m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6e7f9059bac32b018a019e70b01470ab389e3704","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-23 18:33:41.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-60-827860-5/1645639864 : SUCCESS in 16m 53s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":5},{"id":"46636ed976a482e5ef5ee6c4bc602693d0a08df4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-02-23 18:42:29.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/60/827860/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b4ec85f/ : FAILURE in 2h 34m 39s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/60/827860/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/e6e6c87/ : SUCCESS in 52m 54s","accounts_in_message":[],"_revision_number":5},{"id":"b6d691d4b8d0b664597a9e2f30c23840b0b02096","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-23 19:25:14.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/60/827860/5/check/dsvm-tempest-cisco-zonemanager-job/0481125/ : SUCCESS in 1h 09m 34s","accounts_in_message":[],"_revision_number":5},{"id":"8dc7b5aeb1464405ce906b8e26c7a8fc956e9ad4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-02-23 19:40:31.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/60/827860/5/check/iscsi-driver-primera/62b16b8/62b16b8.html : SUCCESS in 1h 45m 36s","accounts_in_message":[],"_revision_number":5},{"id":"2dd249af21b85e2ab309d3067f2cf7acdff67373","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-02-23 19:52:15.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/60/827860/5/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/0e8049e/0e8049e.html : FAILURE in 1h 48m 55s","accounts_in_message":[],"_revision_number":5},{"id":"813e09a117ac294bb376e37b7d68808c8d21a247","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-23 20:06:27.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/60/827860/5/check/iscsi-driver-alletra-6k/8eedce6/8eedce6.html : SUCCESS in 2h 03m 58s","accounts_in_message":[],"_revision_number":5},{"id":"f626e4372cbc2f2fea2d10de2e3e9c122d83923d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-23 20:06:36.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"829ce7967141e365bce7a792e955008610451225","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 20:07:43.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a5b3127585606afef16496f03ba64e05bbb166c7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-23 20:27:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b13fc264691bdd0c8b140c7ba4251e97fec1f702","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 21:03:12.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aad54f990b074e1ca12e5bb9dcc0942a : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e881b9302bd4925b4d52082508d03f7 : FAILURE in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc55238502e247cba8b0a369c4df1dd7 : SUCCESS in 11m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/386471ea9eae48c2899075339b0fab39 : FAILURE in 6m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/ec9c53092e1344cc95e40b230f6acc60 : SUCCESS in 51m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c572a6d67c45444d9caae87667131a07 : FAILURE in 15m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/097457a9463d4876806851d8ed94964d : SUCCESS in 44m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14cbcf9f96854f148bb21f798bb98d23 : SUCCESS in 6m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b6dc3e0eaa7b4cdea8f47b48a13b8b2c : FAILURE in 53m 05s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7aeb38fe37be42f7bcdd2314b35f4296 : POST_FAILURE in 16m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cbbe1241d8e4b25ad2a3a8f1299d6df : FAILURE in 13m 33s","accounts_in_message":[],"_revision_number":5},{"id":"d88c9cbaf120510e99588250239eda3b03bf9152","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-23 21:55:06.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-827860-4/ns5_nfs : SUCCESS in 4h 29m 5s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-827860-4/ns5_iscsi : SUCCESS in 4h 31m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"b453116cf9c4ac713ba98bcb4c6879aea9c9ea5f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-23 21:58:30.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/827860/5/check/tempest-dsvm-full-focal-py3/fcfa1ca/ : SUCCESS in 2h 12m 16s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/827860/5/check/devstack-plugin-ceph-dsvm-focal-py3/d9ce688/ : TIMED_OUT in 3h 14m 51s","accounts_in_message":[],"_revision_number":5},{"id":"fabc351cd5b7f8bd5128e0171e2b15d17ac42da7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-23 22:19:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6b24a93ad704e26a38048e62005e2a9fd343f76c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-23 22:19:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8379ba7ac7c9b4eafd20d332753ce69788823c5a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-23 22:31:26.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/60/827860/5/ : SUCCESS in 1h 43m 01s","accounts_in_message":[],"_revision_number":5},{"id":"5fa077ca936d33f8106e53992e326203a9111a8e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-23 22:33:38.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/827860/5/iscsi/ : FAILURE in 13m 54s\n- cinder-smb http://cloudbase-ci.com/cinder/827860/5/smb/ : FAILURE in 13m 55s","accounts_in_message":[],"_revision_number":5},{"id":"a6b4db6a3643883eae93e8fb3e2eebacae446aa1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 22:43:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a3a19485d064c4f9fef8dfd1a132e9c : SUCCESS in 20m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aacd98ecba6548afa4192d08201715b4 : FAILURE in 21m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d3410cbdc4ec4bb6928968b69ea9ddd9dad6f73c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-23 23:31:59.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-827860-5/ns5_nfs : SUCCESS in 4h 27m 59s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-827860-5/ns5_iscsi : FAILURE in 30m 31s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"6b0c267bea7007d3b9161d945d52d5aa386c8708","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-02-23 23:42:25.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/9dae72b/index.html : SUCCESS in 1h 21m 43s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/8d156a8/index.html : FAILURE in 1h 09m 05s","accounts_in_message":[],"_revision_number":5},{"id":"29dad5aab5ded9f48ee531b4bec506aaf8409466","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-23 23:59:21.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/827860-5/HPMSA-iSCSI/ : SUCCESS in 48:35.30\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/827860-5/HPMSA-FC/ : SUCCESS in 43:55.93","accounts_in_message":[],"_revision_number":5},{"id":"5531faf2547ffbeb64bd963578a0f05aaea55f03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 00:18:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/906eeb7eb9074804a25566adfb63c2f4 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a3e196ef72e4c83b0301c82e61a1286 : SUCCESS in 9m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b38fc5bea5de47d78cac8cde2adca8c2 : SUCCESS in 11m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8cf88c23bc74bdbb054320528c87a19 : SUCCESS in 10m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/d6ae66abc0fe4a98bc864fd7cd7fe70c : SUCCESS in 42m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0ad46c2d8330480ab7da8a2004445ee9 : SUCCESS in 1h 01m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d3a2188f31e47b1b2fba271b7f9544f : SUCCESS in 37m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9da8b93431364254bfc23da017f9af52 : SUCCESS in 6m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3eccbb4f401e4e938b098eb571e05638 : SUCCESS in 14m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dc55252a7f9e463ea0f664bd9d68f725 : SUCCESS in 5m 38s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fc33d348d57248269cfaa570aa245124 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/46062cd0d0a34a8fb7bed5e1fef63e64 : SUCCESS in 6m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ce839ee3d3a1434ea07ea0ff928557fa : SUCCESS in 6m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1a2b3c8b5b7946e998c7efc2af4a5e8c : SUCCESS in 39m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ad7a3923e7e64adbb13721615b126c68 : SUCCESS in 5m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c91dc27dba8b4fa5ae8c4e221509eacf : SUCCESS in 1h 12m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/035caf428a584ba9935f958ffa8a2512 : SUCCESS in 1h 17m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/66b01c3fb98a4580b07cb253bcfaf43c : SUCCESS in 1h 31m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9cb55417705d4922bf32dc3cda0e0ee5 : SUCCESS in 55m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/490e705663c24612ad03a1bc82fdbc27 : SUCCESS in 1h 11m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8631e8377fd64a62baf2cc31d0a1b54f : SUCCESS in 28m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4c1c5ad3fca54704babe85887486e348 : SUCCESS in 1h 10m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d628e3f0fca14ec494048d54447e3d8d : SUCCESS in 1h 55m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef038e13443046298df579400055709e : SUCCESS in 45m 46s","accounts_in_message":[],"_revision_number":5},{"id":"6af9afdc414ff43f70375454fbb3e6ecf3bb41c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 00:19:31.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a3101ea93abac8585515d2363fcfbf396c03bfbb","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-24 00:36:49.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-60-827860-5/1645661925 : SUCCESS in 18m 02s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":5},{"id":"f2b52b4fa2f7147a63fcc0bedc1713be8a17adee","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-24 00:59:18.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/827860/5/check/tempest-dsvm-full-focal-py3/45d9181/ : SUCCESS in 1h 50m 11s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/60/827860/5/check/devstack-plugin-ceph-dsvm-focal-py3/05eba94/ : SUCCESS in 2h 22m 20s","accounts_in_message":[],"_revision_number":5},{"id":"34e7334877e6ffcbca21cfeb34d5b5016cb67051","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 01:31:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcaa1c5b9b28430aaee1ae0ef1b338af : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc69cdf6ad964b52bab13a9c7edcb35d : SUCCESS in 10m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab3b5514d857410b83e6b0d7f1cfcec3 : SUCCESS in 12m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c8ffadbab894849b57234a08160379e : SUCCESS in 10m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/588f95142c7e41a1ac0d22c1f3553404 : SUCCESS in 50m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/71e2ad30956443b1b5eb09269b7eef3c : SUCCESS in 1h 07m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/650cb22b12294845b4e120b4f0b4c65f : SUCCESS in 43m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f26a1ab520b44712b0c5d42914f61193 : SUCCESS in 6m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e02a33275a034b8e99b561d110d039f6 : SUCCESS in 53m 05s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f175f65091d942029476b7d69fb11ab9 : SUCCESS in 1h 10m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b063e84c97c74cb7b5b20b46e0db4c0a : SUCCESS in 42m 12s","accounts_in_message":[],"_revision_number":5},{"id":"b2b2760397af752189992ee502334e39dd95b213","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 01:31:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"ba4638b13bd1b662a5d56f72343b2fbba24c2e43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 01:33:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfd4d5d208ec48978902a8dc0b9ea834 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5495efaf7c3346eeab97ea395b9cb687 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":5},{"id":"1d50b959bcdb30cbc71f26ce58fe19672d0a6691","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-24 02:30:30.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/827860/827860/5/check-not-vote/ext-cinder-zuul/3065e6d : FAILURE in 11m 52s","accounts_in_message":[],"_revision_number":5},{"id":"4dc9d4a4016a4c7f872b9a261e4084ee3614853a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-24 02:32:14.000000000","message":"Patch Set 5:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15313/","accounts_in_message":[],"_revision_number":5},{"id":"ca7d63f78ab526e2b3336ca8ec898533db67c3e6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-24 02:33:59.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15313/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"b80dbe8ff940b908a8f2b32b3948608afd00788a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-24 03:08:38.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/60/827860/5/check/dsvm-tempest-cisco-zonemanager-job/eac0497/ : SUCCESS in 1h 09m 29s","accounts_in_message":[],"_revision_number":5},{"id":"3ebfd149f61123561a9ef62c3873ef6c099b66ed","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-02-24 03:40:54.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/60/827860/5/check/dsvm-tempest-zadara-driver/7bb4128 : SUCCESS in 1h 09m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"27450a7ec64c4a8447acb99d780625a6fa3673f1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-24 06:11:27.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"96e50463666aa75848272e38b93939b1a95b2b2d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-24 06:11:31.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"6c8357e0c706a635e4b1602615d97ab41cedefd7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-24 13:00:06.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-827860-5/ns5_nfs : SUCCESS in 4h 25m 2s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-60-827860-5/ns5_iscsi : SUCCESS in 4h 25m 49s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"a93606266b123bf747705b03d507d1afe97a7863","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-24 18:55:39.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/827860-5/HPMSA-iSCSI/ : SUCCESS in 51:27.26\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/827860-5/HPMSA-FC/ : SUCCESS in 43:21.10","accounts_in_message":[],"_revision_number":5},{"id":"e626b8a188146bac98b5dd2c938b3a6614ad8a3d","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-24 22:57:57.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/827860-5/Lenovo-iSCSI/ : SUCCESS in 1:01:01\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/827860-5/Lenovo-FC/ : SUCCESS in 57:28.73","accounts_in_message":[],"_revision_number":5},{"id":"feb8dc4b88512515a8e72057a776df977741e31f","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-25 08:06:50.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/827860-5/PVME-iSCSI/ : SUCCESS in 1:04:18\n- PVME-FC https://os-logs.seagate.com/dell-ci/827860-5/PVME-FC/ : SUCCESS in 1:00:50","accounts_in_message":[],"_revision_number":5},{"id":"607f0e6d1b788d99f82adc5b96186ca6b0b0ca66","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-02-25 13:12:28.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/827860-5/Seagate-iSCSI/ : SUCCESS in 1:07:01\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/827860-5/Seagate-FC/ : SUCCESS in 1:04:25","accounts_in_message":[],"_revision_number":5},{"id":"3d973566069fe83dc450be624dfafc21840b348a","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-25 16:39:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/827860-5/Lenovo-iSCSI/ : FAILURE in 1:10:27\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/827860-5/Lenovo-FC/ : SUCCESS in 57:13.40","accounts_in_message":[],"_revision_number":5},{"id":"4288d639c05fd54da139886384fc173cf72ec667","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-26 08:09:10.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/827860-5/PVME-iSCSI/ : SUCCESS in 1:06:45\n- PVME-FC https://os-logs.seagate.com/dell-ci/827860-5/PVME-FC/ : SUCCESS in 1:01:24","accounts_in_message":[],"_revision_number":5},{"id":"04efdf218f3da66c19931bffbe68df6d7040d297","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-02-26 13:46:39.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/827860-5/Seagate-iSCSI/ : SUCCESS in 1:06:39\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/827860-5/Seagate-FC/ : SUCCESS in 1:03:47","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"18b8033b6bd1945bd27b03b779a32a883df5fb10","revisions":{"1563ba634a1352b554af1ac02e27518422a2eed4":{"kind":"REWORK","_number":1,"created":"2022-02-04 13:08:36.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/60/827860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/827860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/827860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/827860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/827860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/827860/1"}}},"commit":{"parents":[{"commit":"ae5fd170f0a51306cc9ba44ba9d5fe4169fed239","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae5fd170f0a51306cc9ba44ba9d5fe4169fed239"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-02-04 11:42:25.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-02-04 13:08:06.000000000","tz":60},"subject":"Fix request id","message":"Fix request id\n\nThis patch fixes 3 issues related to request ids:\n\n- Duplicated request ID in logs\n- Missing request id header in response\n- Wrong request ID in response header when using noauth\n\nThe first two are caused by the same issue, which is the fact that\nrequests that don\u0027t require authorization (those that go to / to get\navailable versions) don\u0027t set a context and resuse the previous one.\n\nAnd the last bug is fixed as a freebie when the code in auth.py is\nrefactored to reduce code duplication on setting the cinder.context key\nin the request\u0027s environment.\n\nNot sure if values now defined in ENV_OVERWRITES are still necessary,\nso the patch preserver current behavior.\n\nClose-Bug: #1960019\nClose-Bug: #1960020\nClose-Bug: #1960021\nChange-Id: Ide831a221a7561bbb7da193f0edb011f1989420e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1563ba634a1352b554af1ac02e27518422a2eed4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1563ba634a1352b554af1ac02e27518422a2eed4"}]},"branch":"refs/heads/master"},"d129d04af046a962293aceb64e044a0580417686":{"kind":"REWORK","_number":2,"created":"2022-02-08 15:47:05.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/60/827860/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/827860/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/827860/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/827860/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/827860/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/827860/2"}}},"commit":{"parents":[{"commit":"ae5fd170f0a51306cc9ba44ba9d5fe4169fed239","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae5fd170f0a51306cc9ba44ba9d5fe4169fed239"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-02-04 11:42:25.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-02-08 15:22:48.000000000","tz":60},"subject":"Fix request id mess on Cinder-API","message":"Fix request id mess on Cinder-API\n\nThis patch fixes multiple issues related to request ids:\n\n- Duplicated request ID in logs\n- Missing request id header in response\n- Wrong request ID in response header when using noauth\n\nThe source of these issues is the fact that the request ID and the\nrequest context are set too late, or not at all, in the paster filter\nchain and that it\n\nAnd the last bug is fixed as a freebie when the code in auth.py is\nrefactored to reduce code duplication on setting the cinder.context key\nin the request\u0027s environment.\n\nThe patch fixes these issue by changing the filter chains to have the\nrequest_id filter as the first filter, and using our own filter for it,\nto ensure that the context is correctly set.\n\nNot sure if values now defined in ENV_OVERWRITES are still necessary,\nso the patch preserver current behavior.\n\nCloses-Bug: #1960019\nCloses-Bug: #1960020\nCloses-Bug: #1960021\nCloses-Bug: #1960329\nChange-Id: Ide831a221a7561bbb7da193f0edb011f1989420e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d129d04af046a962293aceb64e044a0580417686"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d129d04af046a962293aceb64e044a0580417686"}]},"branch":"refs/heads/master"},"141383cff9ab99bb49e2602299b909aa8e00a324":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-02-09 12:15:00.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/60/827860/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/827860/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/827860/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/827860/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/827860/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/827860/3"}}},"commit":{"parents":[{"commit":"ae5fd170f0a51306cc9ba44ba9d5fe4169fed239","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae5fd170f0a51306cc9ba44ba9d5fe4169fed239"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-02-04 11:42:25.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-02-09 12:12:52.000000000","tz":60},"subject":"Fix request id mess on Cinder-API","message":"Fix request id mess on Cinder-API\n\nThis patch fixes multiple issues related to request ids:\n\n- Duplicated request ID in logs\n- Missing request id header in response\n- Wrong request ID in response header when using noauth\n\nThe source of these issues is the fact that the request ID and the\nrequest context are set too late, or not at all, in the paste filter\nchain.\n\nAnd the last bug is fixed as a freebie when the code in auth.py is\nrefactored to reduce code duplication on setting the cinder.context key\nin the request\u0027s environment.\n\nThe patch fixes these issue by changing the filter chains to have the\nrequest_id filter as the first filter, and using our own filter for it,\nto ensure that the context is correctly set.\n\nNot sure if values now defined in ENV_OVERWRITES are still necessary,\nso the patch preserves current behavior.\n\nCloses-Bug: #1960019\nCloses-Bug: #1960020\nCloses-Bug: #1960021\nCloses-Bug: #1960329\nChange-Id: Ide831a221a7561bbb7da193f0edb011f1989420e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/141383cff9ab99bb49e2602299b909aa8e00a324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/141383cff9ab99bb49e2602299b909aa8e00a324"}]},"branch":"refs/heads/master"},"e8389f52be7e5eaec366e416cb4fc06ff9536061":{"kind":"REWORK","_number":4,"created":"2022-02-23 15:13:18.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/60/827860/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/827860/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/827860/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/827860/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/827860/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/827860/4"}}},"commit":{"parents":[{"commit":"ae5fd170f0a51306cc9ba44ba9d5fe4169fed239","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae5fd170f0a51306cc9ba44ba9d5fe4169fed239"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-02-04 11:42:25.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-02-23 15:13:09.000000000","tz":60},"subject":"Fix request id mess on Cinder-API","message":"Fix request id mess on Cinder-API\n\nThis patch fixes multiple issues related to request ids:\n\n- Duplicated request ID in logs\n- Missing request id header in response\n- Wrong request ID in response header when using noauth\n\nThe source of these issues is the fact that the request ID and the\nrequest context are set too late, or not at all, in the paste filter\nchain.\n\nAnd the last bug is fixed as a freebie when the code in auth.py is\nrefactored to reduce code duplication on setting the cinder.context key\nin the request\u0027s environment.\n\nThe patch fixes these issue by changing the filter chains to have the\nrequest_id filter as the first filter, and using our own filter for it,\nto ensure that the context is correctly set.\n\nNot sure if values now defined in ENV_OVERWRITES are still necessary,\nso the patch preserves current behavior.\n\nCloses-Bug: #1960019\nCloses-Bug: #1960020\nCloses-Bug: #1960021\nCloses-Bug: #1960329\nChange-Id: Ide831a221a7561bbb7da193f0edb011f1989420e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e8389f52be7e5eaec366e416cb4fc06ff9536061"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e8389f52be7e5eaec366e416cb4fc06ff9536061"}]},"branch":"refs/heads/master"},"18b8033b6bd1945bd27b03b779a32a883df5fb10":{"kind":"REWORK","_number":5,"created":"2022-02-23 15:49:40.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/60/827860/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/827860/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/827860/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/827860/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/827860/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/827860/5"}}},"commit":{"parents":[{"commit":"5b89039e9e51a429c6894e2b8791a3ee4fc338cd","subject":"Merge \"Remove the need for project_id from API endpoints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5b89039e9e51a429c6894e2b8791a3ee4fc338cd"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-02-04 11:42:25.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-02-23 15:48:45.000000000","tz":60},"subject":"Fix request id mess on Cinder-API","message":"Fix request id mess on Cinder-API\n\nThis patch fixes multiple issues related to request ids:\n\n- Duplicated request ID in logs\n- Missing request id header in response\n- Wrong request ID in response header when using noauth\n\nThe source of these issues is the fact that the request ID and the\nrequest context are set too late, or not at all, in the paste filter\nchain.\n\nAnd the last bug is fixed as a freebie when the code in auth.py is\nrefactored to reduce code duplication on setting the cinder.context key\nin the request\u0027s environment.\n\nThe patch fixes these issue by changing the filter chains to have the\nrequest_id filter as the first filter, and using our own filter for it,\nto ensure that the context is correctly set.\n\nNot sure if values now defined in ENV_OVERWRITES are still necessary,\nso the patch preserves current behavior.\n\nCloses-Bug: #1960019\nCloses-Bug: #1960020\nCloses-Bug: #1960021\nCloses-Bug: #1960329\nChange-Id: Ide831a221a7561bbb7da193f0edb011f1989420e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/18b8033b6bd1945bd27b03b779a32a883df5fb10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/18b8033b6bd1945bd27b03b779a32a883df5fb10"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
