)]}'
{"id":"openstack%2Fcinder~830092","triplet_id":"openstack%2Fcinder~master~I3e5fedcda77602f8a97cecf015ebfecf7f5f7fcc","project":"openstack/cinder","branch":"master","topic":"remove-legacyfacade","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-04-27 09:57:25.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-04-27 09:57:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3e5fedcda77602f8a97cecf015ebfecf7f5f7fcc","subject":"db: Migrate \"quota\" APIs to enginefacade","status":"MERGED","created":"2022-02-20 19:04:44.000000000","updated":"2022-05-02 05:08:13.000000000","submitted":"2022-04-27 09:57:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"830092-remove-legacyfacade","meta_rev_id":"6876de4efc7e0314e6ca99ff7fb0fa44bbbe3f8d","_number":830092,"virtual_id_number":830092,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-27 09:57:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-05-02 05:08:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-27 08:25:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-27 08:25:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-05-02 05:08:13.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-20 19:10:18.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-20 20:03:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-20 21:35:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-20 22:06:41.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-02-20 23:08:45.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-02-21 00:37:49.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-02-21 00:58:09.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-02-21 02:41:41.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-21 02:44:45.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-02-21 06:20:27.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-21 16:28:40.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-02-21 17:12:14.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-21 18:54:53.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-21 20:33:04.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-02-21 21:19:11.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-02-21 21:51:38.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-02-22 19:56:09.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-24 08:44:03.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-03-08 14:25:19.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-04-07 21:51:13.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-04-12 10:11:54.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-04-13 07:59:45.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-13 10:47:30.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-04-27 08:25:53.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-05-02 05:08:13.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"4701a09a1cd85b0bf7ff13d84d89e1975b960742","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-20 19:04:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c20ba0817fc244d2c0253b2430ef919a33458937","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-20 19:10:18.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15214/","accounts_in_message":[],"_revision_number":1},{"id":"eb6d256850f8362d3ed5901da7cbcd24fe583d4b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-20 19:12:06.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15214/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"72c896dd970521f6208c0408ae15efcb3a7909df","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-20 19:28:49.000000000","message":"Topic remove-legacyfacade-full removed","accounts_in_message":[],"_revision_number":1},{"id":"bc459d2a55a10054030555a6b06cf4c97c208c2a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-20 19:28:51.000000000","message":"Topic set to remove-legacyfacade","accounts_in_message":[],"_revision_number":1},{"id":"f7e7fa38e760dd5d2c498c33aef5653a9c84f7f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 20:03:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f47d20c093c14cd8b9b72d30e7c29ea9 : FAILURE in 28m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cda6fb73ce8541f5a46f20d5f122e700 : FAILURE in 21m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"995500daafaa0ccfb15bcabd481aaf10f9a026a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-20 21:35:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/033dbae19b5d48f7a59c5cd3bebf9a1f : SUCCESS in 9m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bfdca1bd11f48b9a15872ffbda05e56 : SUCCESS in 11m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b595e926efb14bee9c640386cf4bc9e5 : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ddfc34acd9b4a5489cac954663f9cc0 : SUCCESS in 12m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/0ecc412b85304ac18b538287ba50ae9a : SUCCESS in 1h 01m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a86e402ef6c449fe9844c3992ca539e7 : SUCCESS in 1h 06m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0db5059938a414a9948859c9427b1b2 : SUCCESS in 44m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/04fed17e57974800bf5b7503469275f0 : SUCCESS in 13m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c51b9845d7a34597be271581d45bbf19 : SUCCESS in 6m 47s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/638a55addcb14642b93aaadcc4c6983c : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bebb53880c094327b6fc687e9514b7d9 : SUCCESS in 7m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/33da597b42eb4459a190b95e8e0992ba : SUCCESS in 7m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7e6ea9fc77754f42b6f439a03591c39a : SUCCESS in 42m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c81e9817cbea41b3810e59297415382e : SUCCESS in 6m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/24bab6a53f134ff3b83e76d9c85ae0f2 : SUCCESS in 1h 04m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/822276319d3b4ed68c7118700b4dce3a : SUCCESS in 1h 19m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fbee424cd9734d4980d71e2eb4d28a20 : SUCCESS in 2h 04m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8a3562f8539f4b798e8777401c526939 : SUCCESS in 1h 05m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9225e993499041918e6f2ea8a0bb766e : SUCCESS in 1h 15m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5fc94c46e48f44f2b50ee0043095514c : SUCCESS in 45m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/97d5c3e461db42f5a2246f2ddbd5ac44 : SUCCESS in 1h 13m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9e7ef69b2700474f9d044f4a3d97393f : SUCCESS in 2h 23m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75c1e30368ad406ebe2a4feb5b3d725f : SUCCESS in 46m 17s","accounts_in_message":[],"_revision_number":1},{"id":"dee5cebda54bcb5556ec7c3fee7061ea40ed2d63","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-02-20 22:06:41.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/830092/1/iscsi/ : FAILURE in 54m 24s\n- cinder-smb http://cloudbase-ci.com/cinder/830092/1/smb/ : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":1},{"id":"814aa3d09ed594cbc2779d9e88a4efb73abec4f0","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-20 23:08:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-92-830092-1/1645392910 : SUCCESS in 17m 07s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":1},{"id":"d5597553719256e140a2ec0ea363a59755693417","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-21 00:37:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/92/830092/1/check/dsvm-tempest-cisco-zonemanager-job/3a3ad32/ : SUCCESS in 1h 09m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f80a7f421a87edee93e1a778ee4e5b461210b5f4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-02-21 00:58:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/830092/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/804207e/ : SUCCESS in 1h 41m 22s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/830092/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/5b1f435/ : SUCCESS in 1h 28m 55s","accounts_in_message":[],"_revision_number":1},{"id":"92ab3cb4fe2c7541465d86a0c254e2a2ea7250bd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-21 02:41:41.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/830092/1/check/tempest-dsvm-full-focal-py3/c2fd652/ : SUCCESS in 2h 29m 25s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"295bc1d10911fab4ddb1091b13d6de928ae16cd3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-02-21 02:44:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/92/830092/1/check/iscsi-driver-alletra-6k/a1184d1/a1184d1.html : SUCCESS in 1h 39m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0f341f2005cb3128a92dc6bd6b087105e1803ded","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-21 06:20:27.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-830092-1/ns5_nfs : SUCCESS in 4h 35m 47s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-830092-1/ns5_iscsi : SUCCESS in 4h 35m 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"6ae6169c76ec8f69c2d58e5986fd84f6513d78fe","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-02-21 16:28:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/830092-1/HPMSA-iSCSI/ : SUCCESS in 51:13.89\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/830092-1/HPMSA-FC/ : SUCCESS in 43:52.00","accounts_in_message":[],"_revision_number":1},{"id":"a3cb21dc7e9432b76adf4398ebf73469afc4847f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-02-21 17:12:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/92/830092/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/88ebc9c/88ebc9c.html : SUCCESS in 1h 21m 46s","accounts_in_message":[],"_revision_number":1},{"id":"2c0bef5b029e3efc7d5ed44765b6e68bde3c076a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-02-21 18:54:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/92/830092/1/ : SUCCESS in 1h 41m 55s","accounts_in_message":[],"_revision_number":1},{"id":"37f3135fde079e1c971cd7fc8856838efe510c23","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-02-21 20:33:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/830092-1/Lenovo-iSCSI/ : SUCCESS in 1:02:13\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/830092-1/Lenovo-FC/ : SUCCESS in 56:38.47","accounts_in_message":[],"_revision_number":1},{"id":"c824bdad8191e281eff5bd4a55df43ea2ae2af79","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-02-21 21:19:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/830092-1/PVME-iSCSI/ : SUCCESS in 1:03:18\n- PVME-FC https://os-logs.seagate.com/dell-ci/830092-1/PVME-FC/ : SUCCESS in 1:02:52","accounts_in_message":[],"_revision_number":1},{"id":"592ae0b28d476617f6d662133fe760ede6443b8c","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-02-21 21:51:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/830092-1/Seagate-iSCSI/ : SUCCESS in 1:05:30\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/830092-1/Seagate-FC/ : SUCCESS in 1:01:45","accounts_in_message":[],"_revision_number":1},{"id":"b45730281e15cc7a6c8a79a246c3b8d633bdd4a7","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-02-22 19:56:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/830092/1/20220223013726/NEC-FC : SUCCESS in 3h 18m 07s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/830092/1/20220223013726/NEC-ISCSI : SUCCESS in 3h 18m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c4398ea7f89f9e42bfa754d4ae9ea13338b01283","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-02-24 08:44:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/830092/1/20220224113107/NEC-V-FC : SUCCESS in 2h 02m 51s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/830092/1/20220224113107/NEC-V-ISCSI : SUCCESS in 6h 12m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d09a6fe2c9f1597f547e680e54dd6d0a0ac4675b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-03-08 10:57:58.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"d58e520302a01681478610f9e33389d65f5305f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 12:10:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/36dd8a1a46cf444baa7caf28e4553cdf : FAILURE in 19m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26af956f84d240feb4efdd372db1787a : FAILURE in 27m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7e04ded2c875e99904b79c31f3792188e7574407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 13:46:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/066bf91091ab4b919d7215a76ba511db : SUCCESS in 9m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e63e6a2051904360b301b9c6756f8b9e : SUCCESS in 15m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a3e52c89b8c43b6bf40f69a0b0802df : SUCCESS in 15m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaff84e5aa164bfa80e701890e355541 : SUCCESS in 15m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/d6cc116add424adfa6cfa0d494aa8380 : SUCCESS in 1h 02m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3684d5bfb2e34537bef9e8e2b342a4ab : SUCCESS in 1h 12m 40s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b176c32b99c9442083f783141d2309f1 : SUCCESS in 1h 10m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d2a5d8189db479d89554786cbcf8ffc : SUCCESS in 50m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d3083eb1be854dfe82e2b62ec148daa3 : SUCCESS in 16m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2cd74a68313f4c659781062887c508a3 : SUCCESS in 10m 07s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/41f20f7b8628426cbd158cad0f5ed47c : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f9b208d79bbd4aba9cf7076ae1309f1e : SUCCESS in 9m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8c09392001264654bd3fa3503268d548 : SUCCESS in 10m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c7e275acdbeb4fb5a4a05f8b7d422c74 : SUCCESS in 45m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0c0c4277fc3d4e2b8a1234f1bc15f99c : SUCCESS in 9m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/da49bb881cc440818a412dcf152e0253 : SUCCESS in 1h 34m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/51a4ef9b6eb340ddb58f3efb4b8e2849 : SUCCESS in 1h 35m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/270934cfb19243db9beabddff4c9e7a8 : SUCCESS in 1h 35m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/da7ece8db1c4466bac96114ba5703ab3 : SUCCESS in 1h 07m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7176b326e6204de19df0c73803ab0b1d : SUCCESS in 1h 26m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e261dd69b1f4482197580aa57bf42ee7 : SUCCESS in 56m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8908cfcc121a466a8273cc05c48250ee : SUCCESS in 1h 19m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/89aca3e5ffb44f51964374aa48d46cf5 : SUCCESS in 2h 35m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ae07201b695405d874018561848b590 : SUCCESS in 46m 54s","accounts_in_message":[],"_revision_number":2},{"id":"497ef055d6cef682cc8faa66bfabb37eed3dbfb9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-03-08 14:11:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/830092/2/iscsi/ : SUCCESS in 58m 26s\n- cinder-smb http://cloudbase-ci.com/cinder/830092/2/smb/ : SUCCESS in 1h 13m 20s","accounts_in_message":[],"_revision_number":2},{"id":"e5807c1a80180e876424044c20b6d20b7b5b253a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-03-08 14:25:19.000000000","message":"Patch Set 2:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/08_03_2022_14_16_34\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"46a7e2c4915a2ac2061fbb686600b463837d96aa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-03-08 15:33:14.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/830092/2/check/tempest-dsvm-full-focal-py3/0535a31/ : SUCCESS in 2h 36m 14s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/92/830092/2/check/devstack-plugin-ceph-dsvm-focal-py3/dfccf1a/ : TIMED_OUT in 3h 13m 19s","accounts_in_message":[],"_revision_number":2},{"id":"ef61315ba2e9ba17ec8738a0ce4c237f46d164ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-03-08 16:31:49.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"717ecf8029ec5f0a9a52cc9c7bcf4042b8043689","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-03-08 16:45:22.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"c48ddfcd083a0334a527eabb90509dace85f8e7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 19:22:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba82547fc93c469c913433b28cbc0675 : SUCCESS in 10m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c9fde5cd01d4cdca8121638ad4b38ca : SUCCESS in 13m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb62174770334153b14e42d2083014a5 : SUCCESS in 15m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/523b80702a9a4d268f1bb1e872838870 : SUCCESS in 16m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/4547954067eb4634911d5f3e0d7f6267 : SUCCESS in 1h 04m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9df0548974fb41f8aeadcf33e22e691b : SUCCESS in 1h 05m 28s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c7d4187abf95449f93b0d7835c38657f : SUCCESS in 1h 20m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e84b4332473945b7aa9ca3690fb33b7a : SUCCESS in 52m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/90b40abf451a45f1a864f8a0bd367cfa : SUCCESS in 17m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8ffabea6d62940988a7ecebd9aff8593 : SUCCESS in 8m 29s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b4cbe8eeeada45aaa3f1a0007ebd522e : SUCCESS in 10m 08s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/58d108d0ee464d109aaa64823b74a3c9 : SUCCESS in 8m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0bf13154a3424fa48369c1fd874e89d6 : SUCCESS in 11m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/aa29ea1ee30242699e2ad28bf63c0027 : SUCCESS in 49m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1fad826016384c9883a91154fec546fb : SUCCESS in 10m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1410d8df2092438b82850a0a06e2ed01 : SUCCESS in 1h 16m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0c7a0cb7b1d74c5c88ec44806675275c : SUCCESS in 1h 29m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/14d17c0715fd4b96a3e41f918e5794bd : SUCCESS in 1h 43m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dcdd6b8bea56462ea302b239265bb58b : SUCCESS in 1h 04m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/65a3d603d1b14eedb906bbf12aeb84f0 : SUCCESS in 1h 33m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c5afd6bf5b8b4122ae2f3c544453c9e1 : SUCCESS in 1h 00m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cc42e4ecdee247b7a351ae8d91e5f26d : SUCCESS in 1h 19m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c3f6fdef9fa74dce9e7645240adee01c : SUCCESS in 2h 32m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f99a3f251719472dab75df6acc2954b8 : SUCCESS in 58m 43s","accounts_in_message":[],"_revision_number":3},{"id":"38d7ba3642b4e5438715f00efafe25ce5b903254","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 19:31:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/547e1236d0d340049248b87657b64925 : FAILURE in 24m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd02a7d2b49c44b6aa687c5e4c1ae01a : FAILURE in 22m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"575372ec2f15f06ee8408ec8f611a5eca0159464","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-03-08 19:51:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/830092/3/iscsi/ : SUCCESS in 1h 27m 35s\n- cinder-smb http://cloudbase-ci.com/cinder/830092/3/smb/ : SUCCESS in 57m 36s","accounts_in_message":[],"_revision_number":3},{"id":"4cb199a969ca023034d0c8f1b7ca1e291b48e028","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-08 22:51:51.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-830092-2/ns5_nfs : SUCCESS in 4h 32m 44s\n* NexentaStor5_Cinder_iSCSI http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-92-830092-2/ns5_iscsi : SUCCESS in 4h 38m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"442838ec5a81524c46290b8570cd939675eb9f94","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-09 00:24:21.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://hpe-storage.github.io/hpe-primera-logs/logs/92/830092/3/check/iscsi-driver-primera/fccb0cc/fccb0cc.html : FAILURE in 1h 19m 35s","accounts_in_message":[],"_revision_number":3},{"id":"d2e521d2fbd67c1bc0c173d99cb2b6433203e79f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-09 00:53:05.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/92/830092/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/1a7d324/1a7d324.html : FAILURE in 1h 21m 54s","accounts_in_message":[],"_revision_number":3},{"id":"e44ed46105458a222727b6807471b71b66875f1d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-03-09 01:10:23.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/830092/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4d3203a/ : SUCCESS in 2h 06m 43s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/830092/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/2cff241/ : SUCCESS in 1h 24m 43s","accounts_in_message":[],"_revision_number":3},{"id":"978f67ade7db54866082c9d3b80a90ad58c92254","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-09 01:24:18.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/92/830092/3/check/iscsi-driver-alletra-6k/fa7ea3c/fa7ea3c.html : FAILURE in 1h 36m 50s","accounts_in_message":[],"_revision_number":3},{"id":"ffd870075a45e8ef0ebcb5d84c31401973da19ef","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-03-09 01:46:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/92/830092/3/ : SUCCESS in 1h 31m 59s","accounts_in_message":[],"_revision_number":3},{"id":"8f0ceca6960fc38fbbabb79d4e94a88c4a9e3739","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-09 12:01:00.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://93.100.254.243/nexenta/refs-changes-92-830092-3/ns5_nfs : SUCCESS in 4h 56m 37s\n* NexentaStor5_Cinder_iSCSI http://93.100.254.243/nexenta/refs-changes-92-830092-3/ns5_iscsi : FAILURE in 6h 2m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"a54b5caaa17c711f5915a05d729f5d2f32b96925","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-03-09 20:03:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/830092-2/HPMSA-iSCSI/ : SUCCESS in 50:15.63\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/830092-2/HPMSA-FC/ : SUCCESS in 45:47.90","accounts_in_message":[],"_revision_number":2},{"id":"0e2e0fcc96d4ce816d89051bd32b7078490209b6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-10 02:26:53.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/830092/3/check/iscsi-driver-nimble/15ec4cd/15ec4cd.html : FAILURE in 1h 39m 58s","accounts_in_message":[],"_revision_number":3},{"id":"1bdaec58efdd6dbd8f49336bae39933046231a69","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-03-10 03:56:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/830092-2/Lenovo-iSCSI/ : SUCCESS in 1:02:20\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/830092-2/Lenovo-FC/ : SUCCESS in 58:51.65","accounts_in_message":[],"_revision_number":2},{"id":"9244c12a5d9fbd7488d14c5ecfdcc8fe2820f675","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-10 04:13:33.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/830092/3/check/iscsi-driver-nimble/c63282e/c63282e.html : FAILURE in 1h 44m 10s","accounts_in_message":[],"_revision_number":3},{"id":"fc32982bb475961d1cefba74b90c2f08f3556ced","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-03-10 07:16:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/830092-2/PVME-iSCSI/ : SUCCESS in 1:06:36\n- PVME-FC https://os-logs.seagate.com/dell-ci/830092-2/PVME-FC/ : SUCCESS in 1:03:03","accounts_in_message":[],"_revision_number":2},{"id":"b4a205f0c7faad31bdf0b15732de31beefb2ec7d","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-03-10 09:34:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/830092-2/Seagate-iSCSI/ : SUCCESS in 1:06:31\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/830092-2/Seagate-FC/ : SUCCESS in 1:04:42","accounts_in_message":[],"_revision_number":2},{"id":"073e6afd02890abdc7e3d3abf047a6d18e6912cb","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-03-10 18:21:49.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/830092-3/HPMSA-iSCSI/ : SUCCESS in 58:15.78\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/830092-3/HPMSA-FC/ : SUCCESS in 48:44.93","accounts_in_message":[],"_revision_number":3},{"id":"1b0e442a67733ecaccb064cca57982d1a78b8878","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-03-11 08:20:50.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/830092-3/Lenovo-iSCSI/ : SUCCESS in 1:02:21\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/830092-3/Lenovo-FC/ : SUCCESS in 58:40.63","accounts_in_message":[],"_revision_number":3},{"id":"bf495d2777e7419899a56035ff5041bbd7dbd523","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-03-11 13:07:01.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/830092-3/PVME-iSCSI/ : SUCCESS in 1:04:27\n- PVME-FC https://os-logs.seagate.com/dell-ci/830092-3/PVME-FC/ : SUCCESS in 1:03:14","accounts_in_message":[],"_revision_number":3},{"id":"6978b7363ce61fc1d0f8c9647bfeaf0ecd046f4f","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-03-11 16:15:46.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/830092-3/Seagate-iSCSI/ : SUCCESS in 1:06:00\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/830092-3/Seagate-FC/ : SUCCESS in 1:04:43","accounts_in_message":[],"_revision_number":3},{"id":"9db8547ad1d6da9dbd4b114b039a6a2f76499e41","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-03-12 01:49:48.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/830092/3/20220312073059/NEC-FC : SUCCESS in 3h 18m 21s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/830092/3/20220312073059/NEC-ISCSI : SUCCESS in 3h 18m 46s","accounts_in_message":[],"_revision_number":3},{"id":"9b7057a3c6d915901653422014994928693904d6","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-03-24 09:07:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/830092/3/20220324115617/NEC-V-ISCSI : SUCCESS in 5h 58m 06s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/830092/3/20220324115617/NEC-V-FC : SUCCESS in 6h 10m 52s","accounts_in_message":[],"_revision_number":3},{"id":"758ca373390b8b264962ea99733274b09188c4c5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-04-07 21:51:13.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"51a37f373db7ee9ad716199ccec84c043b37a50c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-12 10:11:10.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"a9ef9f52928547e874282b03e836c6518c454fa5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-12 10:11:54.000000000","message":"Patch Set 4:\n\nBuild Started","accounts_in_message":[],"_revision_number":4},{"id":"456eaabbf5612877450cc5c1ef3b011ac67284fe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-12 10:23:45.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1649758294554 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"e67d92c0c99b6290ad1c2dd9210351148c263ace","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:45:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c29a9b7980c64ceca744148fbba1b1b7 : SUCCESS in 21m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/54e192eceeaa4b9fa5725ea84c32f72d : SUCCESS in 23m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d81759ee02ae00ccb68a6f690997a30936d132a0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-12 12:32:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/830092/4/iscsi/ : SUCCESS in 1h 15m 26s\n- cinder-smb http://cloudbase-ci.com/cinder/830092/4/smb/ : SUCCESS in 1h 02m 46s","accounts_in_message":[],"_revision_number":4},{"id":"4bb8b0153c37168614a8b844e992b25d2572f3db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 12:59:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c15fa2aca5342ac840c66db1eea9be5 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/baae7d4aa717401c862684bd3d6f53c8 : SUCCESS in 10m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/873f11277b684b18b3806e0c20b86f73 : SUCCESS in 12m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e85fed742d094456927424418775892d : RETRY_LIMIT in 3m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80b1eb90aa8c4b29bb9eca44e1106b3c : SUCCESS in 10m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/aa4825dd267948e38898a922558cb15d : SUCCESS in 47m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ceb3913b569940a9b3507bb668a50eec : SUCCESS in 1h 00m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b82aa6588b54d0a8146ece5aad9e4fc : SUCCESS in 40m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ffec8c0db002409993e5d01ea31b51a2 : SUCCESS in 16m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/23b243e9082743e48da2a3d60b692d7d : SUCCESS in 6m 45s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0b1cb3742e7444318220c59712486d3e : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4e73286819354590bc24edd124581821 : SUCCESS in 7m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9ad967e3ae69482ca9e8f3d1791ba224 : SUCCESS in 7m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4466dbc310314fb295fc59e232537d5d : SUCCESS in 41m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e6f7cdc23ac0495eb94d83829640457a : FAILURE in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e860a91705954c889205dac943b7a7ee : SUCCESS in 1h 29m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/61c3183dc4454ee7817327ba30ad0d3b : SUCCESS in 1h 27m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/45aa9aaf0be4459c8b62bd536753e78a : SUCCESS in 1h 34m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0d678b69244249d2a422a99ece7d9795 : SUCCESS in 1h 09m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/61ab09c825a24d09a0ca32d928610a0e : SUCCESS in 1h 25m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/78253b405078411aa6161fb06d15263e : FAILURE in 57m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/49f955f5ad7849a0a84e8344a9e2cab2 : SUCCESS in 1h 20m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fbeba796c51e4d0cb3e3e8bcc6d65a31 : SUCCESS in 2h 34m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cc706a10f6347baa2b175c9730d7d13 : SUCCESS in 47m 39s","accounts_in_message":[],"_revision_number":4},{"id":"c1a8a5678b7d3225a89a843274bc098d6ae5b879","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-12 13:27:53.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/92/830092/4/check/iscsi-driver-alletra-6k/2134617/2134617.html : FAILURE in 1h 35m 42s","accounts_in_message":[],"_revision_number":4},{"id":"3c9cbf414c592119d0e8022ec30219bdb679a040","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-12 14:04:50.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/9654e1ac1d3a4ccb9722fe89bb4cf4c4 : POST_FAILURE in 2h 01m 54s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/2c5bd59711ad46a2aa9ce4896bd1446c : POST_FAILURE in 3h 00m 38s","accounts_in_message":[],"_revision_number":4},{"id":"495dbf0d06b1bfe9cd06870b5ebf2a89d611cb69","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-12 14:29:51.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/830092 : SUCCESS in 1h 21m 35s","accounts_in_message":[],"_revision_number":4},{"id":"cb470e3358eb3328b83a3856047d551d9a315acf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-12 14:38:20.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/830092 : FAILURE in 1h 30m 18s","accounts_in_message":[],"_revision_number":4},{"id":"92e69389b179da48ebdd233c5bff427375367b1c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-12 16:23:43.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/830092/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/de9ab6b/ : TIMED_OUT in 3h 03m 43s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/830092/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/54ac9c9/ : FAILURE in 2h 47m 16s","accounts_in_message":[],"_revision_number":4},{"id":"fc767aa38e4cb95e56e5a60a35c6e5faf4274cd4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-04-12 16:30:33.000000000","message":"Patch Set 4:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/12_04_2022_16_12_14\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"33c3fffc094de921ba902eb4976488ba655a4185","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-12 20:55:22.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/830092/4/check/iscsi-driver-nimble/61cc598/61cc598.html : FAILURE in 23m 43s","accounts_in_message":[],"_revision_number":4},{"id":"112c714afca84c081415371f4f7884f4fdc953f3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-12 22:13:13.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/830092/4/check/iscsi-driver-nimble/4b0fa8e/4b0fa8e.html : FAILURE in 24m 21s","accounts_in_message":[],"_revision_number":4},{"id":"d6dc44163ee703ec4331515b8c23d6da5d740f84","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-12 23:00:12.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/92/830092/4/check/dsvm-tempest-cisco-zonemanager-job/7a30bd8/ : FAILURE in 34m 28s","accounts_in_message":[],"_revision_number":4},{"id":"f9dc52504fe88ad0005558be9231ef0782656caf","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-04-12 23:01:37.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"8142e210a272deb301da34f089fc9cc44bf6c31d","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-04-12 23:27:24.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"b115dee35e4947ef205cb3f06731cb56e2c639ad","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-04-12 23:58:30.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"033ea241ac30d215e257f1f7420f690ffc7b2f91","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-04-13 00:05:50.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"3d1fd5147e6f427d75316686c1f107e65ef90f6d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-04-13 07:59:45.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/92/830092/4/check/DellEMC_PowerFlex/863e0d7/DellEMC_PowerFlex/587 : SUCCESS in 1h 28m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"90cb5a7144006f5a3d5a4ddc9bb4f7437640f310","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-04-13 10:47:30.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/92/830092/4/check/DellEMC_SC_FC/b65d9a2/DellEMC_SC_FC/460 : SUCCESS in 1h 02m 23s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/92/830092/4/check/DellEMC_SC_iSCSI/f5154d9/DellEMC_SC_iSCSI/481 : SUCCESS in 2h 05m 14s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"9365ad2eb1b2c60485194a3f8f5e377289e99157","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-04-15 09:52:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/830092/4/20220415162106/NEC-V-FC : SUCCESS in 1h 56m 34s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/830092/4/20220415162106/NEC-V-ISCSI : SUCCESS in 2h 30m 50s","accounts_in_message":[],"_revision_number":4},{"id":"64debc924de8f98d2d0055992dde1ca3c1dce6dc","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-04-17 19:46:34.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/830092/4/20220418012654/NEC-FC : SUCCESS in 3h 18m 07s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/830092/4/20220418012654/NEC-ISCSI : SUCCESS in 3h 19m 37s","accounts_in_message":[],"_revision_number":4},{"id":"336880a14d86c3f79fca576fe0d1e4fcb3e03cfb","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-04-27 08:25:53.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"86a64f1d1d8e9bc3bc645e2d8486a93e5882dedf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 08:27:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d8b05146f5af014d3e6e44a36f075a91d80b53cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 09:57:25.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a34650a36e344bbb9b7f05ad0ec322d : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ca203c027364fc7938197a2fab68cdc : SUCCESS in 11m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f2e6e28fc154789829f253d1920a007 : SUCCESS in 14m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61ed333bcc4746feba822e7daf45095a : SUCCESS in 12m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/9b1a9deab2a94ac085b1fe6c4bfb3aa1 : SUCCESS in 50m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/53d6274dd7fe491dbc5ea77f3b5f8ce0 : SUCCESS in 1h 25m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a71fc2181b624ed6b3208bcc99da95d2 : SUCCESS in 55m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/33d6f69d1ee540e29e11722461b34105 : SUCCESS in 56m 53s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bed177c4cae5418ab144792d0607ef58 : SUCCESS in 1h 04m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a8dbbf32c164e71b147795af1195233 : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":4},{"id":"fb3e19b4c570b7931810f86d610c9ca81042f28a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 09:57:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"d11963132ccc45cf4aedb39fa1f1c54d00d91318","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 09:59:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20847877b54b436bb916b0be93315da9 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":4},{"id":"d5fc5c869789fe922f67b7b49466b77552df95ec","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-02 05:08:13.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2948d43ddf5346a6a14ae1581594404d : FAILURE in 18m 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"29a43abd589ca8415f26b6890c9e88b4b88a7ae6","revisions":{"7be160c2f330eeed44620c1c582c8354010f2ab7":{"kind":"REWORK","_number":1,"created":"2022-02-20 19:04:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/92/830092/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/830092/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/830092/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/830092/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/830092/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/830092/1"}}},"commit":{"parents":[{"commit":"2ea9dfb88ed3b63b4a4f887669e033d10a29aaea","subject":"db: Migrate \"cluster\" APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ea9dfb88ed3b63b4a4f887669e033d10a29aaea"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-02-14 16:39:31.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-02-19 13:09:51.000000000","tz":0},"subject":"db: Migrate \"quota\" APIs to enginefacade","message":"db: Migrate \"quota\" APIs to enginefacade\n\nMigrate quota-related APIs from the legacy enginefacade to the modern\ncontext-based enginefacade.\n\nWe remove the \u0027require_context\u0027 decorator from inner functions since\nit\u0027s not needed. These aren\u0027t \"public\" APIs.\n\nChange-Id: I3e5fedcda77602f8a97cecf015ebfecf7f5f7fcc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7be160c2f330eeed44620c1c582c8354010f2ab7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7be160c2f330eeed44620c1c582c8354010f2ab7"}]},"branch":"refs/heads/master"},"e8b4565c6bcdcb271cbfbba9903b42706e60880d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-08 10:57:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/92/830092/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/830092/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/830092/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/830092/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/830092/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/830092/2"}}},"commit":{"parents":[{"commit":"bded200586a91c618376baa129afb952e1ce3eef","subject":"db: Migrate \"cluster\" APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bded200586a91c618376baa129afb952e1ce3eef"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-02-14 16:39:31.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-03-08 10:53:50.000000000","tz":0},"subject":"db: Migrate \"quota\" APIs to enginefacade","message":"db: Migrate \"quota\" APIs to enginefacade\n\nMigrate quota-related APIs from the legacy enginefacade to the modern\ncontext-based enginefacade.\n\nWe remove the \u0027require_context\u0027 decorator from inner functions since\nit\u0027s not needed. These aren\u0027t \"public\" APIs.\n\nChange-Id: I3e5fedcda77602f8a97cecf015ebfecf7f5f7fcc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e8b4565c6bcdcb271cbfbba9903b42706e60880d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e8b4565c6bcdcb271cbfbba9903b42706e60880d"}]},"branch":"refs/heads/master"},"1d7703f08a5c18bc47515a7546ff17221b4b2c45":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-03-08 16:31:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/92/830092/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/830092/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/830092/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/830092/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/830092/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/830092/3"}}},"commit":{"parents":[{"commit":"7d4a9a2762d81a58031beb2b6495b8aebf716792","subject":"db: Migrate \"cluster\" APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d4a9a2762d81a58031beb2b6495b8aebf716792"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-02-14 16:39:31.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-03-08 16:22:17.000000000","tz":0},"subject":"db: Migrate \"quota\" APIs to enginefacade","message":"db: Migrate \"quota\" APIs to enginefacade\n\nMigrate quota-related APIs from the legacy enginefacade to the modern\ncontext-based enginefacade.\n\nWe remove the \u0027require_context\u0027 decorator from inner functions since\nit\u0027s not needed. These aren\u0027t \"public\" APIs.\n\nChange-Id: I3e5fedcda77602f8a97cecf015ebfecf7f5f7fcc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d7703f08a5c18bc47515a7546ff17221b4b2c45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d7703f08a5c18bc47515a7546ff17221b4b2c45"}]},"branch":"refs/heads/master"},"29a43abd589ca8415f26b6890c9e88b4b88a7ae6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-04-12 10:11:10.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/92/830092/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/830092/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/830092/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/830092/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/830092/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/830092/4"}}},"commit":{"parents":[{"commit":"bb2e0023d121d2c7075b28788b089d3358ea322e","subject":"db: Migrate \"cluster\" APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bb2e0023d121d2c7075b28788b089d3358ea322e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-02-14 16:39:31.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-04-12 10:03:50.000000000","tz":60},"subject":"db: Migrate \"quota\" APIs to enginefacade","message":"db: Migrate \"quota\" APIs to enginefacade\n\nMigrate quota-related APIs from the legacy enginefacade to the modern\ncontext-based enginefacade.\n\nWe remove the \u0027require_context\u0027 decorator from inner functions since\nit\u0027s not needed. These aren\u0027t \"public\" APIs.\n\nChange-Id: I3e5fedcda77602f8a97cecf015ebfecf7f5f7fcc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/29a43abd589ca8415f26b6890c9e88b4b88a7ae6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/29a43abd589ca8415f26b6890c9e88b4b88a7ae6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
