)]}'
{"id":"openstack%2Fcinder~834172","triplet_id":"openstack%2Fcinder~master~I9e3df907ce2c2e7ddb568a91f5aa21195f68c5e4","project":"openstack/cinder","branch":"master","attention_set":{"33612":{"account":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"last_update":"2022-08-17 13:04:29.000000000","reason":"Someone else replied on the change"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-03-20 15:22:57.000000000","reason":"yuval replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9e3df907ce2c2e7ddb568a91f5aa21195f68c5e4","subject":"lightos - utlize extra_specs_op for compression flag","status":"NEW","created":"2022-03-17 13:59:08.000000000","updated":"2025-10-31 16:00:04.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"bc47ffb0c420a683bcdebcabca1d6b185f7b6361","_number":834172,"virtual_id_number":834172,"owner":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-20 17:56:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-31 16:00:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-17 13:59:18.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-17 14:34:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-17 14:48:22.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-17 15:39:17.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-03-17 16:06:11.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-03-17 16:18:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-17 16:33:42.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-17 17:03:50.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-03-17 17:06:28.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-03-17 17:32:43.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-17 18:07:15.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-03-17 18:25:48.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-19 00:37:45.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-20 17:27:57.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-20 18:08:30.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-03-20 18:10:11.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-03-20 18:10:11.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-03-20 18:16:49.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-03-20 19:33:11.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-03-20 19:58:56.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-03-20 20:30:45.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-03-20 20:52:19.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-21 02:16:06.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-31 16:00:04.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"}],"messages":[{"id":"9db1d3e3b68385b9baec1c2eda1da6ddc21a48f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-03-17 13:59:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20bdeaea12d763c2baab42d229b818731ca09f9f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-17 13:59:18.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15571/","accounts_in_message":[],"_revision_number":1},{"id":"0646c998d23afca2ba633696aac1415bb79282e4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-17 14:01:09.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15571/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d14176cc07c1532157f4a49d3f95e1de5fff5284","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-17 14:24:02.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/834172/834172/1/check-not-vote/ext-cinder-zuul/0e126f8 : FAILURE in 24m 11s","accounts_in_message":[],"_revision_number":1},{"id":"7e34fa9119f47c532642c294510d09a90111fa2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 14:34:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc3bfca483134985ac4b439316d54359 : FAILURE in 31m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/435658e8cab446c3af88682bc7d0fc9d : FAILURE in 32m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db484bad492b5ccc8d264beebb11032945861f85","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-17 14:48:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/834172 : FAILURE in 47m 15s","accounts_in_message":[],"_revision_number":1},{"id":"150fe3fce675a9fba88ad4442309820480bde557","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-17 15:23:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/72/834172/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/d018c63/d018c63.html : SUCCESS in 1h 22m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0adc1474a7033b8c8ba9db25afd2bc5fe10721ee","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-17 15:39:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/72/834172/1/check/iscsi-driver-alletra-6k/2799fe6/2799fe6.html : FAILURE in 1h 38m 56s","accounts_in_message":[],"_revision_number":1},{"id":"abdbb237718696843593ca682e5620a206f7173a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-17 15:39:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/72/834172/1/check/iscsi-driver-nimble/e72d9c4/e72d9c4.html : FAILURE in 1h 38m 53s","accounts_in_message":[],"_revision_number":1},{"id":"89e1217fa27c26216657d2b990c398b7cb9efc01","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-03-17 16:06:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"36fc82a51ca2ff129f8fe876aa2f0452f07f2a3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 16:18:21.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bac70723ebfb4a6f93cdc9b8a5b6c783 : FAILURE in 8m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c807ccf915348a7b2b1198a65a64bcf : FAILURE in 14m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6a11425d2a4404489614884000e8ea7 : FAILURE in 13m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a63126894a0f47ea89d5ff075dd2db40 : RETRY_LIMIT in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bdc60b10bec4272a8556f0bb9e0013c : SUCCESS in 14m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/771cbe9dd7f24ca189224cc1a53b8b5c : SUCCESS in 47m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0171a45048054d33b6fb72cfd4263d4b : SUCCESS in 1h 10m 14s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a328298865f3481e8ef8f4d0d10809d8 : SUCCESS in 1h 17m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58c2b8a7ec8643238bafa2155ad88265 : SUCCESS in 1h 01m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bc50ffe0aa1e43a6aa0068ced4a52623 : FAILURE in 17m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5662f22dfebd467eb539c69f64400d00 : SUCCESS in 8m 23s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c71cb9436519457ca468cb32067d868d : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ea17cb3e807e4cd09486b8ef7ce7ffbf : SUCCESS in 8m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/14a1320af897499a8ea4fef9b5cf58c8 : SUCCESS in 10m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ca0d5c6b61ff40aabcb35d6c5fa0f009 : SUCCESS in 40m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e9a7de91f6f34f70a5e29fda0d9f58b8 : SUCCESS in 8m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d9cacf5aebf74aa1bb4fffd7248c5d28 : FAILURE in 1h 09m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b0e75e0757424efbadd17788245a34d8 : FAILURE in 1h 26m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/17a116ef10424bd596fecef8025586ad : POST_FAILURE in 1h 30m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/efe93c1e760648308e25a2ee6d3a64d0 : SUCCESS in 1h 02m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/29e134a22b99496fb364ba1668ba31f5 : SUCCESS in 1h 24m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fd0ba92668774b31b72aeaaf2dc81348 : SUCCESS in 45m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7bc7e5d0240c4ffb9813ff8a2f81a674 : SUCCESS in 1h 28m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad5df3f7888342caab3baddbb4a21b6d : SUCCESS in 2h 16m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fec059c86a30413cbcd10869d95b9d2e : SUCCESS in 46m 37s","accounts_in_message":[],"_revision_number":1},{"id":"8429d7a5efd0b80272a37dfc1b17dfa11a23a3eb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-03-17 16:33:42.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/f01ad2dc4ed444f99f48a1b91906f270 : POST_FAILURE in 2h 00m 01s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/547eaf778675469581524cdcad29624d : POST_FAILURE in 2h 28m 56s","accounts_in_message":[],"_revision_number":1},{"id":"ccbe72eb51bf1082450722d6072387501f90cf2a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-03-17 17:03:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/72/834172/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f2999cd/ : SUCCESS in 1h 23m 36s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/72/834172/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/177cd3b/ : TIMED_OUT in 3h 02m 43s","accounts_in_message":[],"_revision_number":1},{"id":"777d5ccc83e284247739201b1bdb528b3fe97ca4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-03-17 17:06:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/834172/1/iscsi/ : SUCCESS in 2h 53m 04s\n- cinder-smb http://cloudbase-ci.com/cinder/834172/1/smb/ : SUCCESS in 3h 06m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b68cf457f86e01cf4ef01f3659a861bb1b8f73f1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-17 17:20:39.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/72/834172/1/check/iscsi-driver-nimble/3f37388/3f37388.html : FAILURE in 1h 38m 39s","accounts_in_message":[],"_revision_number":1},{"id":"172f21cf824acf0fe8fb47cad959f25ffb5d5f5b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-03-17 17:32:43.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/72/834172/1/ : SUCCESS in 3h 26m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a9f692fe675eeca69df084f6b66f9ac3723e3b94","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-03-17 18:07:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job finger://zuulserver:17979/d192b37d74d74f1f94165fc234dc3b85 : RETRY_LIMIT in 1h 10m 22s","accounts_in_message":[],"_revision_number":1},{"id":"01cb56c725e715edd743f8da0e82a225755607df","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-17 18:25:48.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://93.100.254.243/nexenta/refs-changes-72-834172-1/ns5_nfs : SUCCESS in 4h 22m 28s\n* NexentaStor5_Cinder_iSCSI http://93.100.254.243/nexenta/refs-changes-72-834172-1/ns5_iscsi : SUCCESS in 4h 26m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"93674ef6137561742a792c3185746ae040dcff28","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-03-19 00:37:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/72/834172/1/20220319061754/NEC-FC : SUCCESS in 3h 15m 01s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/72/834172/1/20220319061754/NEC-ISCSI : SUCCESS in 3h 19m 48s","accounts_in_message":[],"_revision_number":1},{"id":"fa0c7c16d474874a41e68cf29846ab1205c8870b","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-03-20 15:22:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d47b2273e6dae051136f3ad70ecf0d7bf995ef53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-03-20 15:26:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ce0ccf083d59e326deffc007969ce509801ddbe5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-20 15:26:59.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15588/","accounts_in_message":[],"_revision_number":2},{"id":"e6a9666bf2046f0f9272f55a023147e6053ee2cc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-20 15:28:34.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15588/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"531a080913e2e969147ac01e2947b1702d3df05f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-03-20 15:52:17.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"b3e1b73c80afed2e9a987f4e7cfb166ab6fe0642","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-20 15:55:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a52e060edfb04f2984f34ccdf378fe6c : FAILURE in 24m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f96479c51cb249e19be0bf466f8bf5bf : SUCCESS in 26m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"be6e305dc6b13639b5c8bf357521d15702d60941","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-20 16:26:19.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/834172 : FAILURE in 57m 27s","accounts_in_message":[],"_revision_number":2},{"id":"09d8a6f0aadc6b9b718df5ab17f7c4c6a6eab760","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-20 16:31:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/72/834172/2/check/iscsi-driver-nimble/b3bb3bd/b3bb3bd.html : FAILURE in 1h 02m 47s","accounts_in_message":[],"_revision_number":2},{"id":"4213b82ba091a4ae38aba3ab41cc454bb7292504","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-03-20 16:33:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/834172/2/iscsi/ : SUCCESS in 1h 05m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/834172/2/smb/ : SUCCESS in 1h 01m 12s","accounts_in_message":[],"_revision_number":2},{"id":"439207571b52d0cce27bc02fe8b9514b0c1139e1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-20 16:52:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/72/834172/2/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/a05d7cf/a05d7cf.html : SUCCESS in 1h 23m 18s","accounts_in_message":[],"_revision_number":2},{"id":"d05142530d930a109375f6fdefe766d48153a67e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-20 17:11:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/72/834172/2/check/iscsi-driver-alletra-6k/1e0d406/1e0d406.html : FAILURE in 1h 43m 21s","accounts_in_message":[],"_revision_number":2},{"id":"3f9c63a4ce72360120b4bfa97644a949df8f19de","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-03-20 17:16:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/72/834172/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/27e422c/ : SUCCESS in 1h 47m 58s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/72/834172/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/bf5c525/ : SUCCESS in 1h 01m 54s","accounts_in_message":[],"_revision_number":2},{"id":"fedd357213b81e3ad9c156f5a7e12df2a51999a7","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-03-20 17:27:57.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/72/834172/2/check/DellEMC_XtremIO_FC/d52f5bb/DellEMC_XtremIO_FC/810 : SUCCESS in 1h 05m 03s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/72/834172/2/check/DellEMC_XtremIO_iSCSI/452afed/DellEMC_XtremIO_iSCSI/821 : FAILURE in 2h 00m 44s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"d31cd63a2a9b1dbd372c1c78dc2db73677a6bd8e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-20 17:33:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/72/834172/2/check/iscsi-driver-nimble/7bc2417/7bc2417.html : FAILURE in 59m 31s","accounts_in_message":[],"_revision_number":2},{"id":"25accec07aa765a032b89f58b3e8960be18a26fc","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-03-20 17:45:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/72/834172/2/ : SUCCESS in 2h 12m 05s","accounts_in_message":[],"_revision_number":2},{"id":"8fb873f5b70cdfa63ba021d7d9605edf751125b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-20 17:56:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66baf2e02e6149f3b91449c9607604aa : SUCCESS in 9m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/337cf85d7c914fb9a9d61c37a440d1f4 : SUCCESS in 12m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efc647bdb0794c6da89b7eb6111ec405 : SUCCESS in 15m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b05d7f1bbfb44f44a7b5350f26974687 : RETRY_LIMIT in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f63ed4deb73c4acebd0213556c9d7753 : SUCCESS in 13m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/5780a83f83bf47f6a8f508231e37bb05 : SUCCESS in 1h 05m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d0c67af3256f4fe88f03713712d07f0d : SUCCESS in 1h 01m 33s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5312f6cc8cc14984b9247e37eec3d80a : SUCCESS in 1h 14m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4d010e82e984e958eaabe04c5d59b55 : SUCCESS in 49m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/964eb38ff89a439380a6e7afe9039020 : SUCCESS in 17m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c89c673aa1714707b323978c187d2314 : SUCCESS in 8m 19s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fa284d39fae04e68a0e6653884c8d427 : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/82668f3ca65747f9a81a0221682cb70e : SUCCESS in 8m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b3586c763feb482a85ecb20ee611a4cd : SUCCESS in 9m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3e4bdd6d560a49ebbb4a5525635065a4 : SUCCESS in 39m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/76b53d4692834ee6b068686e7be9efee : FAILURE in 7m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6eb8335c007f4fe79124347ee66f84ec : SUCCESS in 1h 18m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/38506f12f50e41c2a8c16914c847d506 : SUCCESS in 1h 36m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1547f9ffdce247cf8ca6a89f20105889 : SUCCESS in 1h 34m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e4656be783894752808fc707f5aa441a : SUCCESS in 1h 04m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6b9f2f8ed47543e293144683412189af : FAILURE in 1h 25m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7d69eb4c6d21400e993135c7b56524c7 : SUCCESS in 47m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/beca2e2e32fc4c8aa34c058601a4555c : SUCCESS in 1h 20m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/99c8661e2f5b4c9abf39a3dfbd190347 : SUCCESS in 2h 24m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdc9766631b7477b98a8673f034dfe63 : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":2},{"id":"6948b49f72c0d34c71cfa67a17c7147b5730d281","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-03-20 18:08:30.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/20_03_2022_17_56_10\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"66435a49e7782d60acff340789a8713a8a2b6921","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-03-20 18:10:11.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"a457a0b9a3f18871246c21564c25cff7d92e960a","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-03-20 18:10:11.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"3ce224c56ddba48bd9faea55acc324a590bceeb6","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-03-20 18:16:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-72-834172-2/1647798960 : SUCCESS in 20m 47s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"4990403ad54d8bf433a53ab61945dd615539b685","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-20 19:07:10.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-72-834172-2/ns5_nfs : SUCCESS in 3h 36m 54s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-72-834172-2/ns5_iscsi : SUCCESS in 3h 39m 53s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"1404bb8c368c67baf1287e1db39c193e7259b310","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-03-20 19:20:22.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job finger://zuulserver:17979/1cad2c7bbe3c4cdbaa9db721741ef612 : RETRY_LIMIT in 1h 09m 02s","accounts_in_message":[],"_revision_number":2},{"id":"af750258d95804356088ed5da333545750754d53","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-03-20 19:33:11.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/72/834172/2/check/DellEMC_PowerStore_iSCSI/38c2d2c/DellEMC_PowerStore_iSCSI/358 : SUCCESS in 1h 36m 51s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/72/834172/2/check/DellEMC_PowerStore_FC/0242228/DellEMC_PowerStore_FC/375 : SUCCESS in 1h 16m 23s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"2894c74ae94ed516559f55b072b61b91ff58d7cd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-20 19:40:41.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://93.100.254.243/nexenta/refs-changes-72-834172-2/ns5_nfs : SUCCESS in 4h 10m 30s\n* NexentaStor5_Cinder_iSCSI http://93.100.254.243/nexenta/refs-changes-72-834172-2/ns5_iscsi : SUCCESS in 4h 13m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"391a7ee3e4ffa0976d38d79678fcc8d06ee22758","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-03-20 19:51:02.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/72/834172/2/check/DellEMC_XtremIO_FC/79b38b6/DellEMC_XtremIO_FC/811 : SUCCESS in 1h 04m 03s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/72/834172/2/check/DellEMC_XtremIO_iSCSI/62f8fd3/DellEMC_XtremIO_iSCSI/822 : FAILURE in 1h 54m 41s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"2e4d87557d25b63076d5b30038e77ed3475b35ca","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-03-20 19:58:56.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/72/834172/2/check/DellEMC_SC_FC/3fc750e/DellEMC_SC_FC/344 : SUCCESS in 1h 00m 43s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/72/834172/2/check/DellEMC_SC_iSCSI/1d1d8d9/DellEMC_SC_iSCSI/366 : SUCCESS in 2h 02m 33s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"105daa210a1f023adea3d190c9d28842cf71b73b","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-03-20 20:30:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/834172-2/PVME-iSCSI/ : SUCCESS in 1:01:37\n- PVME-FC https://os-logs.seagate.com/dell-ci/834172-2/PVME-FC/ : SUCCESS in 1:00:17","accounts_in_message":[],"_revision_number":2},{"id":"5274356e76a75a4cc6212b6dd789041c6dece090","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-03-20 20:52:19.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/72/834172/2/check/DellEMC_VNX_iSCSI/69dca7c/DellEMC_VNX_iSCSI/245 : FAILURE in 2h 55m 27s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/72/834172/2/check/DellEMC_VNX_FC/d33eca8/DellEMC_VNX_FC/262 : FAILURE in 2h 55m 52s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"24519237e1725962f86f25ef9a21aa8c7a4f44d8","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-03-21 02:16:06.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/72/834172/2/check/DellEMC_PowerMAX_FC/6f6ec7a/DellEMC_PowerMAX_FC/358 : FAILURE in 3h 42m 45s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/72/834172/2/check/DellEMC_PowerMAX_iSCSI/c08e8b2/DellEMC_PowerMAX_iSCSI/367 : FAILURE in 8h 19m 38s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"e75c1166b41e49ec959ac55d9dfb191fcae30ec0","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-03-21 08:47:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"716dabc750794ae7a28a529769792d284a713194","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-03-21 09:04:49.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/21_03_2022_08_47_37\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"f797dd8da088598da8b636c112170bec364a030f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-03-21 09:20:55.000000000","message":"Patch Set 2:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/21_03_2022_09_12_58\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"ed533ee249696e48ecfbc0f607a01819b1a84bb7","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-03-22 09:26:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/72/834172/2/20220322102305/NEC-FC : SUCCESS in 2h 55m 26s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/72/834172/2/20220322102305/NEC-ISCSI : SUCCESS in 8h 03m 39s","accounts_in_message":[],"_revision_number":2},{"id":"109d092bb65cbe04379ca73c9e5114326afd1d75","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-05-01 10:28:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e751c164c470d6dfc2ea72466b633156eda88d41","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-08-17 12:56:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9b32ab3d0427edc38e4ae127fd50941a20e10b62","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-17 13:04:29.000000000","message":"Patch Set 2:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/17_08_2022_12_56_49\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"bc47ffb0c420a683bcdebcabca1d6b185f7b6361","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2025-10-31 16:00:04.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7bb87fd753ea5ccea5f5f170b39c7e8842d3d462","revisions":{"9a2eed08c2ed814d2a2a29ffa97d0c414f6f588e":{"kind":"REWORK","_number":1,"created":"2022-03-17 13:59:08.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/72/834172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/72/834172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/834172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/834172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/834172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/72/834172/1"}}},"commit":{"parents":[{"commit":"6d678dcf67f45223b18967aaeb63da115c36cb9e","subject":"Merge \"Add a unit test to keep microversions.py up to date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d678dcf67f45223b18967aaeb63da115c36cb9e"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-03-17 13:50:20.000000000","tz":120},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-03-17 13:58:22.000000000","tz":120},"subject":"lightos - utlize extra_specs_op for compression flag","message":"lightos - utlize extra_specs_op for compression flag\n\nfollowup for: Change-Id: I09fe872eca1b112f67bad7297de5893696f83373\n\nwe remove the parse_extra_spec function and utilize the\nextra_spec_ops module.\n\nSigned-off-by: Yuval Brave  \u003cyuval@lightbitslabs.com\u003e\nChange-Id: I9e3df907ce2c2e7ddb568a91f5aa21195f68c5e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9a2eed08c2ed814d2a2a29ffa97d0c414f6f588e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9a2eed08c2ed814d2a2a29ffa97d0c414f6f588e"}]},"branch":"refs/heads/master"},"7bb87fd753ea5ccea5f5f170b39c7e8842d3d462":{"kind":"REWORK","_number":2,"created":"2022-03-20 15:26:48.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/72/834172/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/72/834172/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/834172/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/834172/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/72/834172/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/72/834172/2"}}},"commit":{"parents":[{"commit":"6d678dcf67f45223b18967aaeb63da115c36cb9e","subject":"Merge \"Add a unit test to keep microversions.py up to date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d678dcf67f45223b18967aaeb63da115c36cb9e"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-03-17 13:50:20.000000000","tz":120},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-03-20 15:24:20.000000000","tz":120},"subject":"lightos - utlize extra_specs_op for compression flag","message":"lightos - utlize extra_specs_op for compression flag\n\nfollowup for: Change-Id: I09fe872eca1b112f67bad7297de5893696f83373\n\nwe remove the parse_extra_spec function and utilize the\nextra_spec_ops module.\n\nSigned-off-by: Yuval Brave  \u003cyuval@lightbitslabs.com\u003e\nChange-Id: I9e3df907ce2c2e7ddb568a91f5aa21195f68c5e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7bb87fd753ea5ccea5f5f170b39c7e8842d3d462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7bb87fd753ea5ccea5f5f170b39c7e8842d3d462"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
