)]}'
{"id":"openstack%2Fcinder~834478","triplet_id":"openstack%2Fcinder~stable%2Fyoga~Ic2da41f3962c1108f974aca952bce3da6d6ac277","project":"openstack/cinder","branch":"stable/yoga","topic":"bug/1909207-stable/yoga","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-03-23 16:24:17.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2022-03-23 14:21:51.000000000","reason":"Eric Harney replied on the change"},"30092":{"account":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"last_update":"2022-03-23 16:24:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic2da41f3962c1108f974aca952bce3da6d6ac277","subject":"Prohibit volume manage to an encrypted volume type","status":"MERGED","created":"2022-03-21 20:53:15.000000000","updated":"2022-03-23 16:27:01.000000000","submitted":"2022-03-23 16:24:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"834478-bug/1909207-stable/yoga","cherry_pick_of_change":768458,"cherry_pick_of_patch_set":20,"meta_rev_id":"80d797a0d1640ad7842ac0a995495400c39a3dec","_number":834478,"virtual_id_number":834478,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-23 16:24:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"recommended":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"all":[{"value":1,"date":"2022-03-23 11:48:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":2,"date":"2022-03-23 14:21:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-23 14:12:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":1,"date":"2022-03-23 15:37:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}],"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-21 20:57:28.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2022-03-21 21:16:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-21 21:49:59.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-21 22:37:01.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-03-21 22:47:07.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-03-21 23:00:28.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-03-22 09:07:03.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-22 18:20:48.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"CC"},{"updated":"2022-03-22 18:20:55.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2022-03-22 21:59:28.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-03-23 02:20:23.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-03-23 03:13:20.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-23 11:48:32.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"}],"messages":[{"id":"e6d5836773514655c22a30be2973eeef54b6f86d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-21 20:53:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8ad6a91230007fa06cf592a2efcac61c2ab96095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 21:16:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54a27ad7e588420d8226133865f87105 : SUCCESS in 9m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c59dca7139164b88be4c6a2a3f2aaddf : FAILURE in 11m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/152eb4efd4814bb1a2f72ba0f4f14b35 : SUCCESS in 14m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fb15adcec89441194b65068cbcf727e : SUCCESS in 14m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e94b1127471a43f988cbc4229301a761 : SUCCESS in 10m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fb8db31ddca740fbbad0d80f7f188f89 : SUCCESS in 17m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1fd9dfe2a5864ae99df09838c2698305 : SUCCESS in 7m 47s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d0229065ee61470999bedca5d5ab0b31 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2492508086c84601841ab7c05296e0df : SUCCESS in 8m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cb19af3481b14d6b9a0995ec27fb0c5e : SUCCESS in 9m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8fc48dec89b5430d9f9a84c27ba7d669 : SUCCESS in 8m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"61af712d23b5920cc2904c7380f90fd4b2007672","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 21:35:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d34f301f909a4e6c8f1f18dd5f4e7c0f : SUCCESS in 24m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/50bb3f658ea948fcae63f946edd56b83 : SUCCESS in 25m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b2960ccdfc056fcdbe978c09c2ba3ddfa44db72d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-21 21:49:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/78/834478/1/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/ac4384a/ac4384a.html : FAILURE in 54m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a880a8cf62529cf3c4ce0bba5c6a909572fd8c2d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-21 21:54:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/834478 : FAILURE in 59m 14s","accounts_in_message":[],"_revision_number":1},{"id":"8b7a5cd1158102dc00f3b73cb176dcd48c90e96e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-21 21:55:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"76b61ae2bb511afb3f402143dfd89aec84603c42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-03-21 22:06:21.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"512e788ec98f573d164d1f43345f4e84e3525e32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 22:28:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b06deda1acec40e38a585d9298393c14 : SUCCESS in 9m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b13df73a4a764aefb6128ff25bf518e0 : SUCCESS in 13m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57ea8e0a985f42c991ae769a0c456896 : SUCCESS in 14m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d1e840cfdc2462a81e1cbf42d8ff8be : SUCCESS in 15m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcb8c91514ed42efad900b5af493a4de : SUCCESS in 11m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/026b291d17294c7db1517e3f4f28c494 : SUCCESS in 19m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8e125b129e0a4761bddc77ffbaf8619d : SUCCESS in 8m 09s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/32eea2aaaf994df8805d3798e05c996a : SUCCESS in 10m 26s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/36c16af81cb0466b8fe5889108e396e6 : SUCCESS in 6m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0caa01ac727e4ff99c54a04e4037248c : SUCCESS in 15m 11s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fbe4729498a741b691258c68c4e52111 : SUCCESS in 13m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"699811720b0e9c71b1bb21ba6a81f84414e73420","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 22:35:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/296f4a14fd3641c2b30ef2b2b1e845ed : SUCCESS in 21m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e942360eb1764d02bb750f7cf7daa38d : FAILURE in 25m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7aadc6c0d024fe8e30cdbf6797204126b2b46a68","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-03-21 22:37:01.000000000","message":"Patch Set 3:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/21_03_2022_22_28_44\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"4a1f01da0a5de173dd96f107dab2996d83bbb3de","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-03-21 22:47:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-78-834478-3/1647901716 : SUCCESS in 18m 29s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":3},{"id":"e2c7e579b293d1f144b1f34affb588ffc7ebaaaa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-21 22:58:25.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/834478 : FAILURE in 50m 35s","accounts_in_message":[],"_revision_number":3},{"id":"a27c2ceb2c2ffb32502be49a8e11df2fbbf4f305","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-21 23:00:28.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/78/834478/3/check/iscsi-driver-alletra-6k/a494bbe/a494bbe.html : FAILURE in 53m 14s","accounts_in_message":[],"_revision_number":3},{"id":"9701018df6e86eaa6f75a9e12ee9a662f6bcaa26","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-21 23:00:50.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/78/834478/3/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/6a8e0bd/6a8e0bd.html : FAILURE in 52m 58s","accounts_in_message":[],"_revision_number":3},{"id":"18ecfaa16b8dd8d252c3cfcdb53e2a51c7b0830c","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-03-21 23:04:10.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-78-834478-3/1647902105 : SUCCESS in 17m 01s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":3},{"id":"00c770f2739353ce63b16a55a4e6a7c115cace28","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-22 03:22:39.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/78/834478/3/check/iscsi-driver-nimble/4dcff32/4dcff32.html : FAILURE in 49m 28s","accounts_in_message":[],"_revision_number":3},{"id":"1cbecfbc7827739ffc4aa0c78180cf68fb728ba7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-22 04:14:12.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/78/834478/3/check/iscsi-driver-nimble/e9e21ab/e9e21ab.html : FAILURE in 50m 04s","accounts_in_message":[],"_revision_number":3},{"id":"211df2b4b4048879e975f3e20bd7346bc2b53bdb","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-03-22 09:07:03.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/78/834478/3/ : SUCCESS in 2h 30m 44s","accounts_in_message":[],"_revision_number":3},{"id":"78aa03abf723a0176d3d787981b4101f4a376ab7","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-03-22 18:20:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"db39b2981e85425485c9e0fa16e2ae5e63f20a03","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-03-22 18:20:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f93940091cbbd065b53ecc8cac700b4291415f66","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-03-22 21:59:28.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job finger://zuulserver:17979/137a57afa7784491a66aa00254dcfa17 : RETRY_LIMIT in 20m 42s","accounts_in_message":[],"_revision_number":3},{"id":"1729c3ed414e1bce10d0761c6b38fe316bab4bcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2022-03-23 01:57:28.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"1ece3a887bbd2354fe65764e9c9c1765a49de25a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 02:19:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3563a3359574e6ea5c30433f2f2d25f : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9bad74c28754d59b2c3589b3367a2c3 : SUCCESS in 13m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37b49442524f4ff9a720c0992d55bd31 : SUCCESS in 16m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1f0dedf6d6e4cdbb5ce693671e210f1 : SUCCESS in 11m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f14c81cc250147b3a1fcb25db6b6133e : SUCCESS in 9m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d216518c3e594ac18157adea376bfef5 : SUCCESS in 15m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/47e165f339d548d180db26cd6bb7d37e : SUCCESS in 6m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/af5302f71a6f49e885962ab8ff394c90 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2f88110eec8d485a80cb0e44cef0ee16 : SUCCESS in 7m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/85b5e22c16ed4c89b993ca0295046bb4 : SUCCESS in 9m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/44d19c41d6844217ab51033d1f59e41a : SUCCESS in 7m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"75848ca16d0c8b6e1b4e630dfb424c0eee3e06c3","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-03-23 02:20:23.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/834478/4/upstream-solidfire-check/cinder-solidfire-iSCSI/1aa6727/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"326fe12eb6b8a89098b30c042aeed84a62ea5277","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-03-23 02:27:21.000000000","message":"Patch Set 4:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/23_03_2022_02_19_07\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"4aa491a9afa310968071c4fd226cb51014dc5573","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 02:32:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1b322259849443d90c7a67537d47602 : SUCCESS in 22m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5bb9614987954d7e8b30c593785e8235 : SUCCESS in 29m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"74ff8665b97c35c7b0a5c66ccf03be5392660856","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-03-23 02:33:01.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/834478/4/upstream-solidfire-check/cinder-solidfire-iSCSI/a2aedef/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"7facd5804c1d586261c7939efd77cb9550c630cb","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-03-23 02:37:31.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-78-834478-4/1648001941 : SUCCESS in 18m 28s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"0e816b4a40a8b98e60f7dbeacd1ac043788c9c47","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-23 02:49:40.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/834478 : FAILURE in 50m 38s","accounts_in_message":[],"_revision_number":4},{"id":"680c69ec438b1849bb48c37916f1ceb1ec387777","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-23 02:54:18.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/78/834478/4/check/iscsi-driver-alletra-6k/293d7d1/293d7d1.html : FAILURE in 55m 52s","accounts_in_message":[],"_revision_number":4},{"id":"9ff0bd6a54fd3ff08d5cb70f89a6d3e33e9432a5","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-03-23 02:54:42.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-78-834478-4/1648002744 : SUCCESS in 17m 09s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"03b8f9ea35951225d964a6b5b5c1c57f45d23b0a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-23 03:13:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/834478/4/upstream-check/cinder-cDOT-iSCSI/871c723/ : SUCCESS in 37m 15s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/834478/4/upstream-check/cinder-cDOT-NFS/8ca0e20/ : SUCCESS in 35m 39s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/834478/4/upstream-check/cinder-cDOT-FCP/84179e4/ : SUCCESS in 53m 18s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"5b39f92a1de2933166616596bfa5e2dd0fff7890","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-23 03:23:23.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://hpe-storage.github.io/hpe_cinder_logs/logs/78/834478/4/check/3par-iscsi-driver-master-client-pip-ssa02-dsvm/29c170a/29c170a.html : FAILURE in 1h 24m 21s","accounts_in_message":[],"_revision_number":4},{"id":"f307bf3ae6b6840d69b9e44487c60c25183042ef","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-23 03:27:19.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/78/834478/4/check/iscsi-driver-nimble/fbccb4f/fbccb4f.html : FAILURE in 49m 02s","accounts_in_message":[],"_revision_number":4},{"id":"611cbf9d32841108b88b0726856eb727d3167e88","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-23 04:18:46.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/78/834478/4/check/iscsi-driver-nimble/f5bd3f1/f5bd3f1.html : FAILURE in 49m 53s","accounts_in_message":[],"_revision_number":4},{"id":"ee27049283fce0827920c3cf2837832320b5918c","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-03-23 05:16:30.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/78/834478/4/ : SUCCESS in 1h 52m 13s","accounts_in_message":[],"_revision_number":4},{"id":"f5adad93fa81dc1c3073cc7a61c702d4d974c47b","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-03-23 11:48:32.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2324efdf5cf6f8807b074f22bb88f3ed56e96556","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-03-23 14:12:49.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4ebe5f32c650babe9d5b65691eef46fa9273a082","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-03-23 14:21:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d34a0bc5689ae49da000a5f869b4776706c816c7","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-03-23 15:37:31.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"05c53dad3fbabad4afe49a5ea55bda613d77fc4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 15:38:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5140a07c782ea226de5e7eef4f7484ddfde38609","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 16:24:17.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d94ce3cafb84166aa6a1c04f6a10346 : SUCCESS in 11m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6655a9f08ccb4916ae273a01e2ae3d5f : SUCCESS in 15m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dda9527b24534f4bac79972719c773f4 : SUCCESS in 14m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d9e4ae919b34e898eecd3401896645b : SUCCESS in 16m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c69f44030fff4f59bcfc2a37118ece87 : SUCCESS in 10m 03s","accounts_in_message":[],"_revision_number":4},{"id":"b9f2f9434133a94bc8266cb14ea5eb0b7589f8fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 16:24:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"80d797a0d1640ad7842ac0a995495400c39a3dec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 16:27:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1a2c2d04fee4261813b317c4e621ebe : SUCCESS in 1m 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/896ed4f0354f4945bb0bb526d34128b2 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9c1aa92b9578c1dc1d63d4d9124d873438d495eb","revisions":{"cde6fed2a9b03117556af50a7c5ae6af4d66db64":{"kind":"REWORK","_number":1,"created":"2022-03-21 20:53:15.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/78/834478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/78/834478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/834478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/834478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/834478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/78/834478/1"}}},"commit":{"parents":[{"commit":"a002f635b2b87e9e191308121ee68304c578851e","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a002f635b2b87e9e191308121ee68304c578851e"}]}],"author":{"name":"Yadiel Xuan(轩艳东)","email":"xuanyandong@inspur.com","date":"2020-12-24 09:20:42.000000000","tz":480},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-03-21 20:53:15.000000000","tz":0},"subject":"Prohibit volume manage to an encrypted volume type","message":"Prohibit volume manage to an encrypted volume type\n\nManaging a volume to an encrypted volume type should not be allowed.\nOne reason is that there is no way for an operator to specify an\nencryption key ID for the volume. Another is that we already don\u0027t\nallow a volume of an encrypted type to be un-managed, so this change\nwill be symmetric.\n\nAlso update and correct the api-ref for this call.\n\nCo-authored-by: Yadiel Xuan(轩艳东) \u003cxuanyandong@inspur.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: Ic2da41f3962c1108f974aca952bce3da6d6ac277\nCloses-bug: #1944577\n(cherry picked from commit 8088dc9580356ca51581c56888c9022848f543b1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cde6fed2a9b03117556af50a7c5ae6af4d66db64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cde6fed2a9b03117556af50a7c5ae6af4d66db64"}]},"branch":"refs/heads/stable/yoga"},"eae96d1e3f4ca224a6b8cd5a361e12e828ed7cab":{"kind":"REWORK","_number":2,"created":"2022-03-21 21:55:58.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/78/834478/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/78/834478/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/834478/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/834478/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/834478/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/78/834478/2"}}},"commit":{"parents":[{"commit":"a002f635b2b87e9e191308121ee68304c578851e","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a002f635b2b87e9e191308121ee68304c578851e"}]}],"author":{"name":"Yadiel Xuan(轩艳东)","email":"xuanyandong@inspur.com","date":"2020-12-24 09:20:42.000000000","tz":480},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-03-21 21:52:40.000000000","tz":-240},"subject":"Prohibit volume manage to an encrypted volume type","message":"Prohibit volume manage to an encrypted volume type\n\nManaging a volume to an encrypted volume type should not be allowed.\nOne reason is that there is no way for an operator to specify an\nencryption key ID for the volume. Another is that we already don\u0027t\nallow a volume of an encrypted type to be un-managed, so this change\nwill be symmetric.\n\nAlso update and correct the api-ref for this call.\n\nCo-authored-by: Yadiel Xuan(轩艳东) \u003cxuanyandong@inspur.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: Ic2da41f3962c1108f974aca952bce3da6d6ac277\nCloses-bug: #1944577\n(cherry picked from commit 8088dc9580356ca51581c56888c9022848f543b1)\nchange:\n  cinder/api/contrib/volume_manage.py - used a feature of\n  unittest.mock that was introduced in py38; rewritten to be\n  compatible with py36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eae96d1e3f4ca224a6b8cd5a361e12e828ed7cab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eae96d1e3f4ca224a6b8cd5a361e12e828ed7cab"}]},"branch":"refs/heads/stable/yoga"},"039b36d7da339a9b97db403dc170f3878469fd4d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-03-21 22:06:21.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/78/834478/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/78/834478/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/834478/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/834478/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/834478/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/78/834478/3"}}},"commit":{"parents":[{"commit":"a002f635b2b87e9e191308121ee68304c578851e","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a002f635b2b87e9e191308121ee68304c578851e"}]}],"author":{"name":"Yadiel Xuan(轩艳东)","email":"xuanyandong@inspur.com","date":"2020-12-24 09:20:42.000000000","tz":480},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-03-21 22:05:10.000000000","tz":-240},"subject":"Prohibit volume manage to an encrypted volume type","message":"Prohibit volume manage to an encrypted volume type\n\nManaging a volume to an encrypted volume type should not be allowed.\nOne reason is that there is no way for an operator to specify an\nencryption key ID for the volume. Another is that we already don\u0027t\nallow a volume of an encrypted type to be un-managed, so this change\nwill be symmetric.\n\nAlso update and correct the api-ref for this call.\n\nCo-authored-by: Yadiel Xuan(轩艳东) \u003cxuanyandong@inspur.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: Ic2da41f3962c1108f974aca952bce3da6d6ac277\nCloses-bug: #1944577\n(cherry picked from commit 8088dc9580356ca51581c56888c9022848f543b1)\nchange:\n  cinder/tests/unit/api/contrib/test_volume_manage.py - used a\n  feature of unittest.mock that was introduced in py38; rewritten to\n  be compatible with py36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/039b36d7da339a9b97db403dc170f3878469fd4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/039b36d7da339a9b97db403dc170f3878469fd4d"}]},"branch":"refs/heads/stable/yoga"},"9c1aa92b9578c1dc1d63d4d9124d873438d495eb":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-03-23 01:57:28.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/78/834478/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/78/834478/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/834478/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/834478/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/834478/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/78/834478/4"}}},"commit":{"parents":[{"commit":"a002f635b2b87e9e191308121ee68304c578851e","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a002f635b2b87e9e191308121ee68304c578851e"}]}],"author":{"name":"Yadiel Xuan(轩艳东)","email":"xuanyandong@inspur.com","date":"2020-12-24 09:20:42.000000000","tz":480},"committer":{"name":"Xuan Yandong","email":"xuanyd@outlook.com","date":"2022-03-23 01:57:28.000000000","tz":0},"subject":"Prohibit volume manage to an encrypted volume type","message":"Prohibit volume manage to an encrypted volume type\n\nManaging a volume to an encrypted volume type should not be allowed.\nOne reason is that there is no way for an operator to specify an\nencryption key ID for the volume. Another is that we already don\u0027t\nallow a volume of an encrypted type to be un-managed, so this change\nwill be symmetric.\n\nAlso update and correct the api-ref for this call.\n\nCo-authored-by: xuan \u003cxuanyd@outlook.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: Ic2da41f3962c1108f974aca952bce3da6d6ac277\nCloses-bug: #1944577\n(cherry picked from commit 8088dc9580356ca51581c56888c9022848f543b1)\nchange:\n  cinder/tests/unit/api/contrib/test_volume_manage.py - used a\n  feature of unittest.mock that was introduced in py38; rewritten to\n  be compatible with py36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9c1aa92b9578c1dc1d63d4d9124d873438d495eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9c1aa92b9578c1dc1d63d4d9124d873438d495eb"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
