)]}'
{"id":"openstack%2Fcinder~835881","triplet_id":"openstack%2Fcinder~master~I89fc0f75d4839fedc4fc771e6d23b9526109cf4a","project":"openstack/cinder","branch":"master","topic":"mypy","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-05-27 15:51:29.000000000","reason":"Brian Rosmaita replied on the change"},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2022-05-24 14:28:08.000000000","reason":"Alan Bishop replied on the change"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2022-05-27 18:32:29.000000000","reason":"Change was submitted"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-05-27 18:32:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I89fc0f75d4839fedc4fc771e6d23b9526109cf4a","subject":"mypy: cinder/volume/volume_types.py","status":"MERGED","created":"2022-03-30 17:25:06.000000000","updated":"2022-05-27 18:38:54.000000000","submitted":"2022-05-27 18:32:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":68,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835881-mypy","meta_rev_id":"adc417ca5304a205d3580c40889eafd3d5cb5f3b","_number":835881,"virtual_id_number":835881,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-27 18:32:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-05-27 15:51:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-05-24 14:28:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-24 13:56:55.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-05-27 15:51:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-30 17:27:10.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-30 18:00:04.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-30 18:33:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-30 18:34:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-30 18:44:12.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-03-30 18:49:29.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-03-30 19:18:31.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-30 19:20:55.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-30 20:40:13.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-03-30 21:55:36.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-03-31 02:34:22.000000000","updated_by":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"reviewer":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"state":"CC"},{"updated":"2022-03-31 11:24:02.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-04-04 04:23:19.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-04 17:10:15.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-06 15:27:13.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-04-06 15:30:14.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-06 15:43:00.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-04-06 16:16:39.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-04-06 16:59:18.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-06 18:18:56.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-04-06 18:53:43.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-04-06 20:11:39.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-04-06 20:28:35.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-04-06 20:50:03.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-04-06 21:57:53.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-08 10:32:04.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-11 15:30:05.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-05 14:30:57.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-05-05 17:48:05.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-05-05 21:02:47.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-06 07:53:06.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-05-09 14:38:49.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2022-05-09 21:01:04.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"CC"},{"updated":"2022-05-10 13:11:52.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-05-10 19:11:13.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-05-17 15:04:30.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2022-05-17 23:42:17.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2022-05-23 18:33:44.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"}],"messages":[{"id":"8a2f8a4031d6cdf33416e403064236773e86f5a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-03-30 17:25:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8a5b2cb31fdee55cd7a7436616cd883fd360faf","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-03-30 17:25:26.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"0f5b72af556c35593d01b7bd34ce586a2c2c60de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-30 17:27:10.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15668/","accounts_in_message":[],"_revision_number":1},{"id":"27f36561101cede0760d9f8115db9d7f3b70dfdb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-03-30 17:28:45.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15668/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"eb63e5bb8b2997ff274640be7a179e0587832b35","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-30 18:00:04.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-835881-1/ns5_nfs : FAILURE in 30m 37s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-835881-1/ns5_iscsi : FAILURE in 31m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"f117317101a1401b48e9acdfb4715f41accc897e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 18:33:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb123c09db864e899d80855f944b2a9a : FAILURE in 12m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2560aec83ca74d1a9e82ff941807bb4c : FAILURE in 15m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eee7a79544c8f70be11ecff529f6a8b0c59996d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 18:34:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67b85c73ddd7421daa8742e405eeff3a : FAILURE in 8m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d91f4bbb0dc4b4b8cebf40cc0566b1e : FAILURE in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93cb89c5466a4cd292d7c6b5c87ce566 : FAILURE in 6m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/119ff18a7d85434e8c404dd94921c962 : RETRY_LIMIT in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3e5ae7d67be4adb8644b0ce56705974 : FAILURE in 6m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/d12e78f47f154210b34cee68795c6e08 : FAILURE in 51m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a7b0ed3f49994f6985c2bd696350b2b0 : FAILURE in 29m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c6f8483e4e942578ad7b1daf9caa3a5 : POST_FAILURE in 35m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3ebe334ad479442abafe42411c691da8 : FAILURE in 5m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c3e8bb663a9b48e0bfc9cbe93e24d938 : SUCCESS in 5m 54s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cc77abd330d9473cb5b493a8f8443ed7 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c638b88cf9204d9eaebcb92a0892fd53 : FAILURE in 5m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ae015d661954714a956bd8026f11730 : FAILURE in 6m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/09bc7a8879d54fe48be50af276ff09de : FAILURE in 34m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/63ceef85ecf44cdebc7fae24d34882e5 : SUCCESS in 5m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d24ffca48e0e498aaa6b474d03d7ed5e : FAILURE in 37m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1736b5b344da4fb4b4bbf479655476f2 : FAILURE in 34m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/30c9861dc3484442a6766172d38496e7 : FAILURE in 32m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/20591e7748004d4ea102142c9a030aca : FAILURE in 1h 01m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/963d4028fa954b3a85b82f157a25f1b4 : FAILURE in 27m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/37023a56ebd6487c9c328a772c8b75d8 : FAILURE in 29m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e41bced1296744998415f88431a397c3 : FAILURE in 32m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/68535b15e3904d938c589f89f998d476 : FAILURE in 31m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12c50307cbe949ff8d608fb054ff83e4 : FAILURE in 29m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e73b6f87b246a79bfa10d1121a5fc1d2429d8066","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-03-30 18:44:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/835881/1/iscsi/ : SUCCESS in 1h 17m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/835881/1/smb/ : SUCCESS in 1h 09m 19s","accounts_in_message":[],"_revision_number":1},{"id":"730e73449b70ff682d5b0faa73a1cb8bd7416b0c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-03-30 18:49:29.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/81/835881/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f726377/ : FAILURE in 26m 23s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/81/835881/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/cc29524/ : FAILURE in 22m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e9fd9863e88b811f0177507b2261438cdc94fa81","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-30 19:18:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/835881 : FAILURE in 53m 03s","accounts_in_message":[],"_revision_number":1},{"id":"df58325636e4128818ca96d1733cdd059dab5d48","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-03-30 19:20:55.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/74c2b03c396f4453b1e4e9182b748481 : POST_FAILURE in 1h 13m 06s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/658cfe8998cb4b9fbafe0fedbc57c134 : POST_FAILURE in 1h 17m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5a92ad10ed8c9bc1c33421829e5ef264ddfcaad0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-30 20:10:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/835881 : SUCCESS in 1h 18m 20s","accounts_in_message":[],"_revision_number":1},{"id":"9885f3ba8354a310ddf5971710a7e8f997c1eea5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-30 20:40:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/81/835881/1/check/iscsi-driver-alletra-6k/2741244/2741244.html : FAILURE in 1h 35m 19s","accounts_in_message":[],"_revision_number":1},{"id":"dbcacf4559425ba8c5793a6217afdf1d62aafda6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-03-30 21:55:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/81/835881/1/check/dsvm-tempest-cisco-zonemanager-job/77cfc36/ : FAILURE in 56m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ced0a4f7b2e57a0913d2210702dca7cf56f9df57","author":{"_account_id":32966,"name":"jiang guodao","email":"jguodao@vmware.com","username":"jguodao"},"date":"2022-03-31 02:34:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f7387482f80bc8dd79d26cae7023c2b9389bbd2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-03-31 11:24:02.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/1/20220331193130/NEC-V-ISCSI : FAILURE in 49m 48s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/1/20220331193130/NEC-V-FC : FAILURE in 52m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2d4ae608de3061f2c1f50ddb99f37b098c0053f2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-31 11:34:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/1/check/iscsi-driver-nimble/fe493e0/fe493e0.html : FAILURE in 1h 01m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a6e52a560d6f856a98c00834999bc7df6325e7c6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-03-31 12:41:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/1/check/iscsi-driver-nimble/5ccb97c/5ccb97c.html : FAILURE in 1h 05m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a333f9ee04d0aa15d0e1ca639b3921f4ce31e202","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-03-31 21:54:46.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d6c7bec0d66ab5deb210eeb98271d16c44ee0c94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 22:15:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8550c1cac9b9497584b64b8fe50c283e : FAILURE in 8m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e6f21437bf39422c9e57dbf7ca09a263 : FAILURE in 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c245eac41d59c15e9e8a4a64890cccdedfcabb3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-03-31 22:20:54.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"a3adbbd8bed52c5096c272875e43f24bc0398984","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-31 22:29:41.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-835881-2/ns5_nfs : FAILURE in 30m 9s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-835881-2/ns5_iscsi : FAILURE in 31m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"2e61fd8df89b51930d756807864d93cba75b1833","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 23:07:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0438b09a8ac64624a230cbb18479690c : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3eab0fbadff845498711f2803d3d8ea7 : FAILURE in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be01dab4fe184d6fbb1329d00731ecfe : FAILURE in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56bf6155e035466face9a629fe6cd392 : RETRY_LIMIT in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad6396a9861e4e2783069cb4576967a0 : FAILURE in 6m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/2c6255dc2b474be79cbf378734a4bb07 : FAILURE in 42m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44cf6f12a97441328171ccb99e18715d : FAILURE in 33m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/903730f554484d32bbd3a7212b2d2b49 : POST_FAILURE in 24m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f09313bae52a4f7988fcf8207cde184f : FAILURE in 5m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1a5bd1a113a64976a10c774516bd3116 : SUCCESS in 6m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8a3d87dbdfec468d8316b9a55857da7e : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/88cc9b84eb234542b8c97dedd8a1326a : FAILURE in 6m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2df9a1daee474f7c9bd8cae6972053bf : FAILURE in 7m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cf3d7480f2d74a3b8a3d1c02ed829930 : FAILURE in 32m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d5af13bcaf8b46f7b94b5a2c2b870192 : SUCCESS in 5m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b6ad50c37f174157b996f0957c98d68a : FAILURE in 33m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/85785e0a838947b8ab76192fddbb20f4 : FAILURE in 36m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c0b2b834fbca40b5b0f8925728fa802d : FAILURE in 32m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9394b38e68784002adb1d1aa2b421bce : FAILURE in 1h 03m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1d6a4b08527a4e50a5ddd6786bf7f5d5 : FAILURE in 30m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/038167565c78466e9a9809c8763f5d38 : FAILURE in 28m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2cc4eb1b50a94dd69237eef2ab1f0b41 : FAILURE in 32m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6f1e62115a0148f6b7d9f065c00127b6 : FAILURE in 35m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3999c0a88fb543ad80681429e9aa994d : FAILURE in 23m 41s","accounts_in_message":[],"_revision_number":2},{"id":"6b6855473b0410acf93ac2fee897dc63eaa8efba","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-03-31 23:43:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/835881 : FAILURE in 53m 16s","accounts_in_message":[],"_revision_number":2},{"id":"60debef383dde041bb94e5a7cf3470a3d56fefe1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-01 02:24:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/835881 : SUCCESS in 1h 18m 05s","accounts_in_message":[],"_revision_number":2},{"id":"d829810d1ef49a8f2c585d6d0a103470280eea64","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-01 02:51:20.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/81/835881/2/check/iscsi-driver-alletra-6k/3e6e1c2/3e6e1c2.html : FAILURE in 1h 02m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a862a644ec4467d4cd081b5e121ba4e84f185465","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-01 03:03:57.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/835881/2/iscsi/ : FAILURE in 5h 07m 29s\n- cinder-smb http://cloudbase-ci.com/cinder/835881/2/smb/ : FAILURE in 3h 25m 22s","accounts_in_message":[],"_revision_number":2},{"id":"c08fc3fd47cbae7f9def2e5171520b374ef3c53c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-01 03:27:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/81/835881/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/97a295d/ : FAILURE in 22m 59s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/81/835881/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/27ed6d9/ : FAILURE in 26m 59s","accounts_in_message":[],"_revision_number":2},{"id":"2b2fe59697326f665abea5cb7e8067a67a4cb0af","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-01 12:04:38.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/81/835881/2/check/dsvm-tempest-cisco-zonemanager-job/969d987/ : FAILURE in 56m 16s","accounts_in_message":[],"_revision_number":2},{"id":"7791ea2965fbf57e716bf7117fa40190628e8f45","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-01 18:36:39.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/2/check/iscsi-driver-nimble/6b1367f/6b1367f.html : FAILURE in 1h 04m 36s","accounts_in_message":[],"_revision_number":2},{"id":"bb9f39ea8659b7238019804ea932841ec46f3af2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-01 19:38:38.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/2/check/iscsi-driver-nimble/26a736a/26a736a.html : FAILURE in 1h 00m 13s","accounts_in_message":[],"_revision_number":2},{"id":"9209e7f0a45e119a246a1048870eecb0202f0dc3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-04-04 04:23:19.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/2/20220404123539/NEC-FC : FAILURE in 43m 24s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/2/20220404123539/NEC-ISCSI : FAILURE in 47m 37s","accounts_in_message":[],"_revision_number":2},{"id":"ba42c5b51b87645eced30b0ce8092b4fc7de2e6e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-04-04 13:30:47.000000000","message":"Topic set to mypy","accounts_in_message":[],"_revision_number":2},{"id":"0529eb2e360613b3c51c6e5d61c3a1224178d86b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-04-04 14:49:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"11648176562af3db060d6c20f74bbaa14c67cebf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 15:17:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/07cff1397f11498bb1be2429056f5445 : FAILURE in 20m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a16c4ad85b5e4921a78d426b5bdfd1e1 : FAILURE in 22m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"15b393decea341a913819df015e9691906490af7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-04 15:46:31.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/835881 : FAILURE in 54m 26s","accounts_in_message":[],"_revision_number":3},{"id":"e80ad49676d7e1151606daafa7cc075341d11f84","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-04 15:53:25.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/81/835881/3/check/iscsi-driver-alletra-6k/72deec1/72deec1.html : FAILURE in 1h 01m 55s","accounts_in_message":[],"_revision_number":3},{"id":"ebd5e28e390c8cff58ed5ee1b266b4540cb5df39","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-04 16:14:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/81/835881/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/510628a/ : SUCCESS in 1h 22m 46s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/81/835881/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/2003adc/ : SUCCESS in 1h 07m 32s","accounts_in_message":[],"_revision_number":3},{"id":"498b48892f562e8b782b0ea21f851272d765c355","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-04 16:16:18.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/835881 : SUCCESS in 1h 22m 57s","accounts_in_message":[],"_revision_number":3},{"id":"456e34d5f7e7821619507b0b0763b0e8ddbe7a5b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-04 16:22:59.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/81/835881/3/check/dsvm-tempest-cisco-zonemanager-job/9e21afd/ : SUCCESS in 1h 12m 28s","accounts_in_message":[],"_revision_number":3},{"id":"26467751eb0591d31ad4c720b15a7b6073025756","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-04 16:23:34.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/835881/3/iscsi/ : FAILURE in 1h 32m 53s\n- cinder-smb http://cloudbase-ci.com/cinder/835881/3/smb/ : SUCCESS in 1h 06m 15s","accounts_in_message":[],"_revision_number":3},{"id":"2b38ee40ddb6fae02c9caa6ffb277830fb3528d9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-04-04 17:10:15.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/22def01dd48b48efa5244d37803d9433 : SUCCESS in 1h 19m 08s","accounts_in_message":[],"_revision_number":3},{"id":"18b0023407582223c0ed27ba5edf8533235c3acb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 17:17:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3724a28d07c74d1e8c12d8a43eafd58e : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d302c64ae3c34744b87ef0c6fc8589c5 : FAILURE in 11m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/824886e8355643698a644f4d1670a4bd : FAILURE in 12m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fe215dba6464b708af294510e503d1b : RETRY_LIMIT in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/213e363d05f447e38ad09da813d3be3d : SUCCESS in 9m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/3cbb0bbf104744f5a6c6e78f7e986a2b : SUCCESS in 59m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ec907fafe7847aea4affa3f863e9868 : SUCCESS in 58m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/026fb9f295cc41a2bd1aedccb8b9f204 : SUCCESS in 55m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/019caa2c501a45a6bd93ce5feb5e151b : FAILURE in 12m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e3491b7ba52d47219d5718da5d7ffa00 : SUCCESS in 5m 03s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9dc2e7041f1b456ca1e6dce9f888e24b : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/48424713dafa43f0a28827a2feb95288 : SUCCESS in 6m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/acb0c87799264c84bac379263607b079 : SUCCESS in 6m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/681417f6563e4ce1a6553fe6e567a445 : SUCCESS in 41m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c5430515bac948588797a5ab9ff3460e : SUCCESS in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ee424385f6c84625938a7680452aae4f : SUCCESS in 1h 24m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/749d55654b35412ca97ab933f2d5f868 : SUCCESS in 1h 19m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e9b3aa7163b74e368e14ee42f5d920cb : SUCCESS in 1h 29m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5e5b1047a1854dbfb0b8354e44c40700 : SUCCESS in 1h 14m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4af99451ec4b471f9768ce1d99aa51a9 : SUCCESS in 1h 29m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2a9e662cfb76400aa6b0b4e1303659f6 : SUCCESS in 53m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/119f065698f5426ebe159c750b10323c : SUCCESS in 1h 09m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad4a559b8f544652af845f90cf09fbce : SUCCESS in 2h 19m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17da63e031eb43f882b382dd618cee1e : SUCCESS in 57m 22s","accounts_in_message":[],"_revision_number":3},{"id":"f94e3eb5e9864fcf8a6a2017e412184a982db42c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-04 17:21:49.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/32a5b4e954744f87ae9c55281c335bef : POST_FAILURE in 1h 42m 28s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/65702bdad09a4bbd8a66ef8734010f0d : POST_FAILURE in 2h 28m 30s","accounts_in_message":[],"_revision_number":3},{"id":"a35f8f55b6456c4b4cc2eb6588447daaf1cd4faf","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-04 18:14:22.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-835881-3/ns5_nfs : SUCCESS in 3h 17m 33s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-835881-3/ns5_iscsi : SUCCESS in 3h 24m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"a13a4b008852c67b57714fcaee4e3652941bda4b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-04 18:22:33.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/3/check/iscsi-driver-nimble/f43dede/f43dede.html : FAILURE in 1h 35m 41s","accounts_in_message":[],"_revision_number":3},{"id":"6b56d0e5c140813e253e5b37cfba486e0cbf5a4c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-04 20:02:54.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/3/check/iscsi-driver-nimble/e6f8cf0/e6f8cf0.html : FAILURE in 1h 37m 28s","accounts_in_message":[],"_revision_number":3},{"id":"320d9029bb0e9cbdd3f3d9a37c5f9ec8143e829c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-04-04 23:07:22.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/3/20220405044611/NEC-FC : SUCCESS in 3h 20m 12s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/3/20220405044611/NEC-ISCSI : SUCCESS in 3h 21m 07s","accounts_in_message":[],"_revision_number":3},{"id":"f62d1569621ffcbe555a3d6e10679ac1e7a70904","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-04-05 21:47:05.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"285ffc5036f1ab1737557da68908546e9e988d42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 22:13:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/596c143daf55423086a1608252deb37a : FAILURE in 21m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f8ca87c65dd24d77badf0d70d6102d9c : FAILURE in 23m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b05f7975ce95cf784868c145f4afa37a3158b43f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-05 22:45:17.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/835881 : FAILURE in 55m 54s","accounts_in_message":[],"_revision_number":4},{"id":"759aa474db47d299e2bb351ada1caa3c1f0cad2b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-05 22:50:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/835881/4/iscsi/ : SUCCESS in 1h 03m 02s\n- cinder-smb http://cloudbase-ci.com/cinder/835881/4/smb/ : SUCCESS in 55m 02s","accounts_in_message":[],"_revision_number":4},{"id":"8fd490722cdbe4760683e822fcfb33e010fbdaac","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-05 22:51:48.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/81/835881/4/check/iscsi-driver-alletra-6k/c65850c/c65850c.html : FAILURE in 1h 03m 07s","accounts_in_message":[],"_revision_number":4},{"id":"510e01f7edffce27b1ce4dd38d8b6c9e3978a70c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-05 23:12:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/835881 : SUCCESS in 1h 22m 14s","accounts_in_message":[],"_revision_number":4},{"id":"0f0fd8fd8616a378d2c1e3aeada4b7eeb991b46a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 00:01:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50fd14896360478d883673cd4f62ac1f : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8ef58598039486083d7d9b12001d825 : FAILURE in 9m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db75091f95a14edb849fcde327c48c81 : FAILURE in 13m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43b256d4d0b94de89f09ebadd76b2d1a : RETRY_LIMIT in 2m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76dc95832710473d981e5627151564c6 : SUCCESS in 12m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/f0280fcb78ea43dca17c8276640fdfee : SUCCESS in 55m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc173f47246948a1928f1ae61e36c845 : SUCCESS in 1h 06m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd9f41130eaa4f87b539233365bd250e : SUCCESS in 56m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7972bcebade242eb8a4a95e7268b23d5 : FAILURE in 12m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1fca008a62e04cfc85ca71a8fa225cb5 : SUCCESS in 5m 53s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2ab97636874041a6b82f0fc3478daa85 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cd47b36cd76740eab588c3c2ce7237c7 : SUCCESS in 6m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6ad0cd11f31c4ceb86225f84cb9b3681 : SUCCESS in 7m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5bda13ae05014c5eb2c03d7b4d13eb0b : SUCCESS in 36m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ba2613d4a81e4ddba8e06221f0a3edbd : FAILURE in 4m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b5c1a42168014a71be38eb62ccd5d589 : SUCCESS in 1h 14m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b4a12b1ab4cb4e6daa92bdff7a3df65e : SUCCESS in 1h 23m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/75c6fc731bca4a42a36f8a176457671c : SUCCESS in 1h 30m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ebdd9bfe5d314a31a6d0c41a0d95f4f8 : SUCCESS in 1h 05m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1cd19a44b30547e59070221ec8037d61 : SUCCESS in 1h 15m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ff21133758a145f8878bfada82c2583c : SUCCESS in 41m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4354ad113c2f4a1e93dfc6224cf66aaa : SUCCESS in 1h 11m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/334f5f489a3c4e8db68414390795c703 : SUCCESS in 2h 12m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b903eba8d70f4ee8953ece6a7007afc9 : SUCCESS in 45m 29s","accounts_in_message":[],"_revision_number":4},{"id":"66c63847ed994401bc876a419240663069e95fc3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-06 00:17:29.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/81/835881/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b2b4183/ : FAILURE in 2h 28m 15s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/81/835881/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d81a429/ : SUCCESS in 1h 02m 25s","accounts_in_message":[],"_revision_number":4},{"id":"54247e9fbfe13a49cb78b6e879e33faf6a6516af","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-06 00:51:11.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/cf4d02e4807649ffbe06fa80a60f3567 : POST_FAILURE in 2h 02m 22s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/9449a9e44dc5451398cc9df9884312b7 : POST_FAILURE in 2h 56m 19s","accounts_in_message":[],"_revision_number":4},{"id":"773d9ca3ee426b91912780f5ded4477d680d8931","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-06 01:14:14.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-835881-4/ns5_nfs : SUCCESS in 3h 19m 27s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-835881-4/ns5_iscsi : SUCCESS in 3h 25m 33s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"5d8985704ca21e650c88b1486a04d46b55cf3e81","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-04-06 01:25:59.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/54336eb9d730445bbb1013c94d6d869c : FAILURE in 42m 01s","accounts_in_message":[],"_revision_number":4},{"id":"9424183780548b8590f8b613468bf7d62f78798d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-06 01:51:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/81/835881/4/check/dsvm-tempest-cisco-zonemanager-job/1509537/ : SUCCESS in 1h 12m 38s","accounts_in_message":[],"_revision_number":4},{"id":"b9701312e6485c9b3a3b5e77fb78c7d9a5a1cbb6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-06 07:59:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/4/check/iscsi-driver-nimble/902d113/902d113.html : FAILURE in 1h 36m 39s","accounts_in_message":[],"_revision_number":4},{"id":"088d31e36bd98eb14141de10a7d1299df9101c8b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-06 09:38:43.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/4/check/iscsi-driver-nimble/d0c811e/d0c811e.html : FAILURE in 1h 37m 29s","accounts_in_message":[],"_revision_number":4},{"id":"702d8b4a6448703f0bbceaf568f0a9df4e7a7248","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-04-06 13:08:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ea496bb367ec6cfd62e29d5750ef0ceb3ee3befe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 13:32:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6de94b4d4bd43388b25d844a82bbc5c : SUCCESS in 20m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/27b85130f2f1428eb03bbccadb21caf9 : SUCCESS in 21m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cb43c4862cb06f537e59123920fabe7a3b412dd4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-06 13:43:20.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-835881-5/ns5_nfs : FAILURE in 34m 5s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-835881-5/ns5_iscsi : FAILURE in 33m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"7acd2c1ad655b2c69edafc64df5e9aa30aa0fc52","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-06 14:30:33.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/835881/835881/5/check-not-vote/ext-cinder-zuul/3133ed5 : SUCCESS in 41m 28s","accounts_in_message":[],"_revision_number":5},{"id":"5322eff7044073c18e258421387ce511d2585983","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-06 14:42:50.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/835881/5/iscsi/ : SUCCESS in 1h 29m 03s\n- cinder-smb http://cloudbase-ci.com/cinder/835881/5/smb/ : SUCCESS in 1h 09m 25s","accounts_in_message":[],"_revision_number":5},{"id":"847f971de24c2eb1a477bddfcfc52113dc3858c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 15:25:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07cddcdbfd1741129941841c62abc8c0 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/986aac9014974fb3ba6a920b45b251ad : SUCCESS in 10m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7671a00bab247ffb7583a0875fe0995 : SUCCESS in 11m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19b8c54fa1bc43cbb890717b01c66fcd : RETRY_LIMIT in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a12f2841358b436baed64af37216aafa : SUCCESS in 13m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/cb195eafdbcb461c8ff3bcead69f663a : SUCCESS in 1h 02m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/17a1d76dedca48ab991b087657b65e61 : SUCCESS in 1h 27m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75ee88eb25854dd2a1eb7064f64899b8 : SUCCESS in 58m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fff0705d20d142b89df1d6ce5d7a76bd : SUCCESS in 13m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/37de3f11479449199f6d294f648b5bd1 : SUCCESS in 5m 49s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fc713ad9e0a84909a3429da4c77e3763 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d77c19291f3d400896fc51313f9bf7a6 : SUCCESS in 6m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f2bf6573d3c04e80b853fdbcadcc09f7 : SUCCESS in 8m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/55a86fed3f8b437a821459a285345237 : SUCCESS in 51m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6b5947ae9f6347779a438deede1c9305 : FAILURE in 4m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/485bb37c98ae4ae08b86efcc1540972d : SUCCESS in 1h 22m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1b5680eca3dd43e99643d4717b097524 : SUCCESS in 1h 09m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a31dea9abff642168791136ec8cac45d : SUCCESS in 2h 09m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f65cf7a9d92b433ebf1fd4e237ab2888 : SUCCESS in 55m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a5913dbece0842aabba31291cd9cfcc7 : SUCCESS in 1h 35m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0b9c72901a76441b82734e6d2897b1e4 : SUCCESS in 52m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c0b1bf6eae3f4b64834feaeb5e3fa666 : SUCCESS in 1h 18m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7df34690910c492c8377a8acee1efa3c : SUCCESS in 2h 08m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4245a2934e4f4de1bfbd9ae3acd50f0e : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":5},{"id":"afdf0dd9347a429d0c0a011ae422bf8ae74b1f48","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-04-06 15:27:13.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/5/upstream-solidfire-check/cinder-solidfire-iSCSI/a3a46e6/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"76185a50c3929ce9d4572bf3b8a2ff0d4acad54d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-04-06 15:30:14.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/5/upstream-check/cinder-cDOT-iSCSI/4b93c43/ : FAILURE in 2m 25s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/5/upstream-check/cinder-cDOT-NFS/46f4cac/ : FAILURE in 3m 44s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/5/upstream-check/cinder-cDOT-FCP/e599efc/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"07088e3bb73974d3eb1d1418f11e13de2e3d524c","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-04-06 15:43:00.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-81-835881-5/1649258706 : SUCCESS in 17m 52s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":5},{"id":"ff591a57c2d9828dd386bc224242655cca2ec32d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-04-06 16:16:39.000000000","message":"Patch Set 5:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/06_04_2022_15_46_40\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"224b4f4be7365ffc1e4448b70329f9452e49583a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-06 16:21:53.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/835881 : FAILURE in 57m 26s","accounts_in_message":[],"_revision_number":5},{"id":"2ddfc218a1bf4a1c179307dfb4e13662ed1a6a86","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-06 16:57:26.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/81/835881/5/check/iscsi-driver-alletra-6k/b497410/b497410.html : FAILURE in 1h 06m 31s","accounts_in_message":[],"_revision_number":5},{"id":"c05ec55f4a3f87c7a07a3722550680a1d590e327","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-04-06 16:59:18.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/81/835881/5/check/DellEMC_PowerFlex/8734d00/DellEMC_PowerFlex/535 : SUCCESS in 1h 33m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"5efe7cbce65dc32e49e51870ce38892c5541fbf6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-06 17:40:25.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/835881 : SUCCESS in 1h 19m 18s","accounts_in_message":[],"_revision_number":5},{"id":"678073cbcc41dd066ac2c1ec99cf1426f8911723","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-06 18:16:04.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/8f94bfcf8ce94187a11cc48f4eaea371 : POST_FAILURE in 3h 01m 46s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/ece2718a8ad148ffa0ab81bc3cb6bf7f : TIMED_OUT in 3h 10m 39s","accounts_in_message":[],"_revision_number":5},{"id":"a7dc721dad96463e07eb24f7b2eb8bf8c4d18f05","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-04-06 18:18:56.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/81/835881/5/check/DellEMC_SC_FC/f6243ea/DellEMC_SC_FC/403 : SUCCESS in 59m 58s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/5/check/DellEMC_SC_iSCSI/85df413/DellEMC_SC_iSCSI/424 : SUCCESS in 1h 57m 58s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"d1ea2c884ed193e5738fef1c4e54168f8992b269","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-06 18:48:26.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-835881-5/ns5_nfs : FAILURE in 15m 54s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-835881-5/ns5_iscsi : ABORTED in 1h 44m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"439ced32b9e3a205047a23efa53340d9e97e1361","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-04-06 18:53:43.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/835881-5/HPMSA-iSCSI/ : SUCCESS in 54:27.38\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/835881-5/HPMSA-FC/ : SUCCESS in 45:33.95","accounts_in_message":[],"_revision_number":5},{"id":"c3ca225b5d91800597032c9737243663da5800f3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-06 19:01:42.000000000","message":"Patch Set 5:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15733/","accounts_in_message":[],"_revision_number":5},{"id":"73ce0d121bd1009fbf4fc035d6a67ffb8789666d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-06 19:03:39.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15733/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"2d79145af3ea0f83e73a872466f9220bf6a3882a","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-04-06 20:11:39.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/835881-5/Lenovo-iSCSI/ : SUCCESS in 1:04:24\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/835881-5/Lenovo-FC/ : SUCCESS in 59:03.48","accounts_in_message":[],"_revision_number":5},{"id":"f2276708803b7f98b1ecae9d8f517877712025fb","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-04-06 20:28:35.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/835881-5/PVME-iSCSI/ : SUCCESS in 1:06:41\n- PVME-FC https://os-logs.seagate.com/dell-ci/835881-5/PVME-FC/ : SUCCESS in 1:03:17","accounts_in_message":[],"_revision_number":5},{"id":"c228cd5e2e54ad7c28d49f97cb831d31fe81cc43","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-04-06 20:50:03.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/835881-5/Seagate-iSCSI/ : SUCCESS in 1:07:38\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/835881-5/Seagate-FC/ : SUCCESS in 1:07:47","accounts_in_message":[],"_revision_number":5},{"id":"d33c996a6d6cc3cfa9b8ff5051ea5bde33eaf94f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-06 21:21:51.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/81/835881/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/e866b03/ : FAILURE in 2h 54m 49s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/81/835881/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ecc4868/ : SUCCESS in 1h 25m 56s","accounts_in_message":[],"_revision_number":5},{"id":"53ab0efe38e1c129857854e081bcd35c34e182f1","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-04-06 21:57:53.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/81/835881/5/check/DellEMC_PowerMAX_FC/daf9e18/DellEMC_PowerMAX_FC/420 : FAILURE in 3h 12m 53s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/5/check/DellEMC_PowerMAX_iSCSI/670b69b/DellEMC_PowerMAX_iSCSI/426 : FAILURE in 3h 22m 44s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"597d8d1028da634159bd4c6cfe9891129caab3f7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-04-07 00:40:18.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/affd07edad834ca38b52c525f690380a : SUCCESS in 1h 23m 42s","accounts_in_message":[],"_revision_number":5},{"id":"365e9d0d15facfef596c4b3fdc147876d7b4c9eb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-07 01:10:55.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/81/835881/5/check/dsvm-tempest-cisco-zonemanager-job/52cb55b/ : SUCCESS in 1h 12m 10s","accounts_in_message":[],"_revision_number":5},{"id":"b47f8098e374b9d3166fb1eabf8cd83f40f92b33","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-07 01:16:50.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-835881-5/ns5_nfs : SUCCESS in 3h 28m 55s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-835881-5/ns5_iscsi : SUCCESS in 3h 25m 31s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"ea9d774f0c391f9d04e021f5677c783b4984ee58","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-07 11:59:26.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/5/check/iscsi-driver-nimble/c3fc2f4/c3fc2f4.html : FAILURE in 1h 00m 43s","accounts_in_message":[],"_revision_number":5},{"id":"dc2c32cb25d1c684546428e0495722816c01a22c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-07 13:03:47.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/5/check/iscsi-driver-nimble/ed597b9/ed597b9.html : FAILURE in 1h 02m 32s","accounts_in_message":[],"_revision_number":5},{"id":"691ddc5e6610139874ad70aa640a0c4f9a53b4eb","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-04-08 02:24:27.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/5/20220408085611/NEC-V-FC : SUCCESS in 1h 59m 19s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/5/20220408085611/NEC-V-ISCSI : SUCCESS in 2h 28m 12s","accounts_in_message":[],"_revision_number":5},{"id":"f44090ef02252b0f667498b7228b9fd6073551af","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-04-08 08:17:23.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/5/20220408110043/NEC-ISCSI : SUCCESS in 3h 27m 08s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/5/20220408110043/NEC-FC : SUCCESS in 6h 16m 37s","accounts_in_message":[],"_revision_number":5},{"id":"fcf908ab3e8ef576252e00840077aaa53b09d104","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-04-08 10:32:04.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/81/835881/5/ : SUCCESS in 2h 31m 00s","accounts_in_message":[],"_revision_number":5},{"id":"8c91a074bb5b25f31e35b8c9adcc5748900c5407","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-04-11 15:30:05.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/81/835881/5/check/DellEMC_XtremIO_FC/54f39d1/DellEMC_XtremIO_FC/928 : FAILURE in 14m 50s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/5/check/DellEMC_XtremIO_iSCSI/7411f1c/DellEMC_XtremIO_iSCSI/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"d8637d79246f07d1aa5c0ab93ad3deb894922000","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-05 14:30:45.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4fe3e7e42022e2775993a097c907388214f50a68","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-05 14:30:55.000000000","message":"Patch Set 6:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15953/","accounts_in_message":[],"_revision_number":6},{"id":"8a217136beb866e33cb0f9286a7bebd51e73ae17","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-05 14:30:57.000000000","message":"Patch Set 6:\n\nBuild Started","accounts_in_message":[],"_revision_number":6},{"id":"fe5edf299d2b1edd09e9d2b8996400f93ae12975","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-05 14:32:33.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15953/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"79632f87bfda75d40a18607b86e448cdbc9bf873","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-05 14:35:54.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"856a39b638451f94fb155a8091aa5ae5f5b50482","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-05 14:48:45.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1651761055864 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"479f850cb0a91c00f015924ff97889cc76567551","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 14:55:49.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec732b2260d440a8a7509d6266248178 : SUCCESS in 17m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/93298704b1394b0ba37c3e2284f97df9 : SUCCESS in 18m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b1d23feb967c4167736c85b905e08a69d8972f02","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-05 15:50:39.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/835881/6/iscsi/ : SUCCESS in 1h 16m 46s\n- cinder-smb http://cloudbase-ci.com/cinder/835881/6/smb/ : SUCCESS in 1h 18m 27s","accounts_in_message":[],"_revision_number":6},{"id":"23bbb05c8ba2fd81bb930588fd79cc43fabb8672","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-05 15:54:46.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/835881 : SUCCESS in 1h 21m 48s","accounts_in_message":[],"_revision_number":6},{"id":"4201ab7e4c8e41e0a13c622106429e61d33a6ed7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-05 16:14:00.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/6/check/iscsi-driver-nimble/5d2c934/5d2c934.html : FAILURE in 59m 51s","accounts_in_message":[],"_revision_number":6},{"id":"22ec5bc063e5037bb9ff2992f350d689f1aa89fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 17:00:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/053d8b803a374811b49c8b4892eb9b6b : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c14e34a540d41db95d28db241c66c8f : SUCCESS in 10m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/661157bf96614f3b856938ed9fe5ec36 : SUCCESS in 11m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6ec83750b4c414b93b7c2349646fe93 : RETRY_LIMIT in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d3e8ae8d00a471caafeb7e56b628cfa : SUCCESS in 11m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/1907dbb0f4fa44d09d52026129514a8d : SUCCESS in 1h 06m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2e9159e9f7b34c08b78c7ab2873df78c : SUCCESS in 1h 22m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62606b77729043a080721c3fdfe0755f : SUCCESS in 43m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5f9836a32dbe4d5fa473beb79e4b7771 : SUCCESS in 13m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4c96b50c008d435bac2de6f2b56b3aac : SUCCESS in 5m 35s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b106124ed9504bd69f8274f0c7189e22 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e96524f2d01e4ac2be0b0a9815107158 : SUCCESS in 7m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5f7c36bfac8e4a0cbf2e6a807e8ead14 : SUCCESS in 6m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dd6e66cc60324eac83da81e1495c8a01 : SUCCESS in 34m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/38b64b3bfc1a46cd831fc5b75f404fdf : FAILURE in 7m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/862da1a687c948a19acacea238b032fa : SUCCESS in 1h 29m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1fafcf8dca194e9dbf7644d4e5ba46b3 : FAILURE in 1h 03m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/07483e55a7ce411ea77ffa68d85d6104 : SUCCESS in 1h 32m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b09762e85b274f90a2b24cff1e025a12 : SUCCESS in 1h 07m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/392f839138444cb7a11b3d9c8821c33a : SUCCESS in 1h 38m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/aa3123deacb249f3b8bb753324640453 : SUCCESS in 53m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/185f6e069c0f48d089447e6632811216 : SUCCESS in 1h 30m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b6cf8cbb1009461d8e10ccbf5d6a2571 : SUCCESS in 2h 22m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a3ab7665c4a48efa8953afd9c4ed919 : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":6},{"id":"ddf475186cf6ac675638782e5e4298518a334872","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-05 17:10:08.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/835881/6/check/tempest-dsvm-full-focal-py3/b425d99/ : SUCCESS in 1h 53m 05s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/835881/6/check/devstack-plugin-ceph-dsvm-focal-py3/57d738a/ : SUCCESS in 2h 20m 56s","accounts_in_message":[],"_revision_number":6},{"id":"56aeb59a172c85462508f5d3cf4ed59a6cfbfa34","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-05 17:35:59.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/81/835881/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/32a7bb8/ : TIMED_OUT in 3h 03m 12s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/81/835881/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/c341a4c/ : FAILURE in 2h 55m 52s","accounts_in_message":[],"_revision_number":6},{"id":"3c33038045346a95fc34e873f0ea9b4a68723dbe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-05 17:48:05.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/81/835881/6/check/DellEMC_Unity_FC/5235d15/DellEMC_Unity_FC/735 : SUCCESS in 1h 12m 14s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/6/check/DellEMC_Unity_iSCSI/00f71d8/DellEMC_Unity_iSCSI/653 : SUCCESS in 1h 41m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"920b4fd1d00cc056cf251b5145882d0582750b5d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-05 18:21:20.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-835881-6/ns5_nfs : SUCCESS in 3h 24m 3s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-835881-6/ns5_iscsi : SUCCESS in 3h 26m 16s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"259c11dde51e6b65c14171ec00db01370a4b4c45","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-05 18:34:50.000000000","message":"Patch Set 6:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/05_05_2022_17_56_19\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"b932503e587e545a4316b68b990aaeb39d1286f2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-05 19:14:01.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/81/835881/6/check/DellEMC_PowerFlex/438e070/DellEMC_PowerFlex/699 : SUCCESS in 1h 14m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"24b57b07b9bdec72b2c09347fc767e89a0962c77","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-05-05 20:35:30.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/81/835881/6/check/DellEMC_SC_FC/728c5f0/DellEMC_SC_FC/571 : SUCCESS in 1h 04m 36s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/6/check/DellEMC_SC_iSCSI/a3e34ef/DellEMC_SC_iSCSI/592 : SUCCESS in 1h 57m 52s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"dedf2bcb5062aaf7685a2836444dbf529e39dd87","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-05 21:02:47.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/6/check/DellEMC_VNX_iSCSI/9b1ae43/DellEMC_VNX_iSCSI/359 : SUCCESS in 2h 06m 20s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/81/835881/6/check/DellEMC_VNX_FC/715d799/DellEMC_VNX_FC/376 : FAILURE in 10m 41s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"b2369cdc3bb093ac77562d0ebd30d489ee6947d2","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-05-05 21:26:32.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/835881-6/HPMSA-iSCSI/ : SUCCESS in 53:54.61\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/835881-6/HPMSA-FC/ : SUCCESS in 50:56.85","accounts_in_message":[],"_revision_number":6},{"id":"4e982132229b96a3e375fdf2e7f7d2c132dde354","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-05-05 22:24:27.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/835881-6/PVME-iSCSI/ : SUCCESS in 1:09:51\n- PVME-FC https://os-logs.seagate.com/dell-ci/835881-6/PVME-FC/ : SUCCESS in 1:05:35","accounts_in_message":[],"_revision_number":6},{"id":"e0894bc138ed159a685dd3b6da750675806057ce","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-05-05 22:26:00.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/835881-6/Seagate-iSCSI/ : SUCCESS in 1:09:08\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/835881-6/Seagate-FC/ : SUCCESS in 1:04:55","accounts_in_message":[],"_revision_number":6},{"id":"cdc6a8b8262440bb947c02e02f3ff313f1cfdf0e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-05-06 07:53:06.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/6/check/DellEMC_PowerStore_iSCSI/c303831/DellEMC_PowerStore_iSCSI/484 : SUCCESS in 1h 17m 27s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/81/835881/6/check/DellEMC_PowerStore_FC/c3c0dce/DellEMC_PowerStore_FC/509 : SUCCESS in 1h 11m 52s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"e14779133958a430e1c9a448679f0e611044984d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-06 11:15:35.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3cc04bccddba4bed8af179ed1a7458b5 : SUCCESS in 1h 24m 45s","accounts_in_message":[],"_revision_number":6},{"id":"18f45fc65c1550b066853314a201a684d851b745","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-06 19:34:30.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/6/20220506221952/NEC-V-FC : SUCCESS in 1h 52m 35s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/6/20220506221952/NEC-V-ISCSI : SUCCESS in 6h 14m 34s","accounts_in_message":[],"_revision_number":6},{"id":"313291e2a28fb573862ae1e13ad0993e4f8d62ec","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-05-07 19:18:22.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/6/20220507215549/NEC-FC : SUCCESS in 3h 16m 18s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/6/20220507215549/NEC-ISCSI : SUCCESS in 6h 22m 29s","accounts_in_message":[],"_revision_number":6},{"id":"e48e2d7f2f5a3338a167fb713de95d9791f54ca1","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-07 22:02:48.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/835881-6/Lenovo-iSCSI/ : SUCCESS in 1:04:22\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/835881-6/Lenovo-FC/ : SUCCESS in 58:30.45","accounts_in_message":[],"_revision_number":6},{"id":"b9023eea58268e185f581fba64d30605f48fb275","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-09 14:01:14.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"3473799229280b1a06f669d056f66be569b168f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-09 14:22:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b86640001405387e1434ff04be81f7c8b61a169d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-09 14:22:32.000000000","message":"Patch Set 7:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15959/","accounts_in_message":[],"_revision_number":7},{"id":"01346e40efc0326bb8755a372ffe85b367841f05","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-09 14:22:33.000000000","message":"Patch Set 7:\n\nBuild Started","accounts_in_message":[],"_revision_number":7},{"id":"3922b13f16ca87e257f6fd71086ac6faecfeabea","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-09 14:24:12.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15959/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"fe6646ba59d3f96704e6895ad3d63a83075ce346","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-05-09 14:38:49.000000000","message":"Patch Set 7: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"5efa8593b8a1e6c23b4017310c70e43a62592a0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 14:47:28.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6b844a740e84d8db49041787483a391 : FAILURE in 10m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d4054b21b517476cb8d27cab2d86b92f : SUCCESS in 22m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8339d67847d714dd9c43c3b39fa2344256672149","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-09 14:50:52.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-835881-7/ns5_nfs : FAILURE in 28m 0s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-835881-7/ns5_iscsi : FAILURE in 25m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"23899a1f31dd91b42183f16576cfc569fd240e75","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-09 14:51:15.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/81/835881/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5d970f1/ : FAILURE in 26m 50s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/81/835881/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d0721c7/ : FAILURE in 27m 01s","accounts_in_message":[],"_revision_number":7},{"id":"7bfee3df56955d9d4132114ab49a91afd33f47d4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-09 15:10:11.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/648d79baa4a24bf5bed7c99e15b90e8b : FAILURE in 45m 43s","accounts_in_message":[],"_revision_number":7},{"id":"bb8a1ee6bb4ea7e4f0101f00d329433a1384b7c7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-09 15:16:00.000000000","message":"Patch Set 7:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1652106150215 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"5c94c4cee1b255150e9d4fb4e9d45a1642e96856","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-09 15:17:11.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/835881 : FAILURE in 52m 28s","accounts_in_message":[],"_revision_number":7},{"id":"200d9737dfdd79cc4d405f28df487961e17f9364","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-09 15:18:40.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/81/835881/7/check/DellEMC_Unity_FC/f000c64/DellEMC_Unity_FC/746 : FAILURE in 55m 58s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/7/check/DellEMC_Unity_iSCSI/441de40/DellEMC_Unity_iSCSI/664 : FAILURE in 32m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"00ea466ebbe70cc0e507c8f45e40e0940a4e4870","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-09 15:19:43.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/835881 : FAILURE in 55m 18s","accounts_in_message":[],"_revision_number":7},{"id":"76a0ff34c00d9edecd897a7eeb7e4c4cc77065ed","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-09 15:21:49.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/7/check/iscsi-driver-nimble/6599471/6599471.html : FAILURE in 58m 05s","accounts_in_message":[],"_revision_number":7},{"id":"2e59a0f490f4805ec54d18374e28f2a3f3d73242","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 15:27:05.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f05acb930af40ea8ef19432925c1439 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dae58b735624cb2bad89d995b74f2cc : FAILURE in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dead9f26c65e44a99ce0bad71e42a21b : SUCCESS in 10m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6390cbc0f72348189fb2026af34d7a61 : SUCCESS in 11m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f210fdd7825d4781814d7256dc55c166 : FAILURE in 6m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/4f02669cf5e142dc92d2b0dc416a12aa : FAILURE in 44m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d11e42c91def46d1b257c0840c45f304 : FAILURE in 24m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/911da0e856494c1bb8bd8976baf3af99 : POST_FAILURE in 25m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c72e327e91ab4eeaa8812d5c9c51784d : FAILURE in 5m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/302dea8e9af946d59eb15a9152a40197 : FAILURE in 5m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a8260c6dde64597bd4d4f4dd013c8f7 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8705e7962d464a3ca0453785716b1aab : FAILURE in 5m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/114a5db749fb407b971ad756ca0ef948 : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/195d8e4856ec4c4094275394d5b9f3f0 : FAILURE in 27m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a3ee802985b4483da4c739cd704b7b40 : FAILURE in 6m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f3bf1cb2850342f88287a257788e599e : FAILURE in 27m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e8d537e963fe43f6b245b49e18aa6665 : FAILURE in 38m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ac93bc42b02940e6823ec3d2f22186c5 : FAILURE in 26m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/225ff2311b8c42d6a1a0743a097aaf02 : FAILURE in 59m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1106bdc3a4bc42e1ad70fc51de73a25e : FAILURE in 23m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d109bf7733254cf7bc94b8a17cab3197 : FAILURE in 29m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/649545969f484ad8815aeaf3714fb5a1 : FAILURE in 36m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e049593720924f6ba1983f9515a7cfcf : FAILURE in 30m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6152dc3b8ea74eda9c4ebf67f8581eb7 : FAILURE in 29m 52s","accounts_in_message":[],"_revision_number":7},{"id":"571c90ca5138f150b1c2b97bbb33f2934b82701e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-09 16:23:58.000000000","message":"Patch Set 7:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/7/20220510002923/NEC-V-ISCSI : FAILURE in 49m 59s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/7/20220510002923/NEC-V-FC : FAILURE in 54m 32s","accounts_in_message":[],"_revision_number":7},{"id":"444e4ae66dc21e0aa8a103e8189326bf890fa818","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-09 16:34:58.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/835881/7/iscsi/ : FAILURE in 2h 12m 03s\n- cinder-smb http://cloudbase-ci.com/cinder/835881/7/smb/ : FAILURE in 2h 08m 00s","accounts_in_message":[],"_revision_number":7},{"id":"9b4a0f02e53f1fa803ddc01fca88c9fdc1f4ad74","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-09 16:35:36.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/835881/7/check/tempest-dsvm-full-focal-py3/378343e/ : FAILURE in 2h 10m 57s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/835881/7/check/devstack-plugin-ceph-dsvm-focal-py3/c1c7262/ : FAILURE in 1h 25m 38s","accounts_in_message":[],"_revision_number":7},{"id":"aa412917293dd2b1a45d2f0a278cc5658b64825d","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-09 16:50:55.000000000","message":"Patch Set 7:\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a7d4ac31c51d5cfc7e8b9ca157791921fe39117e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-09 17:25:23.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4a55249057110d41acca454cb401bb4b6afa7f38","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-09 17:25:31.000000000","message":"Patch Set 8:\n\nBuild Started","accounts_in_message":[],"_revision_number":8},{"id":"4b54f4b5ae8fba478816fca0a05d32677dc7c0b9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-09 17:25:34.000000000","message":"Patch Set 8:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15962/","accounts_in_message":[],"_revision_number":8},{"id":"4e65d7201beabcd014249008826d2029409eae14","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-09 17:27:26.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15962/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"244807df96734d7f67f26713a7c9b990b36605f5","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-09 17:42:08.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1fa67451d9e7e30de9388ae2fb83049c1665b60f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-09 17:43:33.000000000","message":"Patch Set 8:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1652117130570 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"097e2110c8743c3656e44fb9286334fa2653e2fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 17:58:19.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37cb0158715948cba18d8d4989a1bb7b : SUCCESS in 24m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/45c49acb42b14ab5bf5190363ebbc7fe : SUCCESS in 24m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0f91fa4e1846f695e6eaf161cb9116763853df92","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-09 18:27:45.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/8/check/iscsi-driver-nimble/7bad08b/7bad08b.html : FAILURE in 1h 01m 02s","accounts_in_message":[],"_revision_number":8},{"id":"71fe638aaf5c6e09bd33d1cb1ab8284cc80db6e6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-09 18:43:42.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/835881 : FAILURE in 1h 16m 09s","accounts_in_message":[],"_revision_number":8},{"id":"fbb5a88f14225bcf74b3048a28e88dd3f19151be","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-09 18:47:02.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/835881 : FAILURE in 1h 19m 53s","accounts_in_message":[],"_revision_number":8},{"id":"00290686d65d3c0ca93aca056f8862dc3b84c9d3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-09 18:52:30.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/835881/8/iscsi/ : SUCCESS in 1h 26m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/835881/8/smb/ : SUCCESS in 1h 17m 04s","accounts_in_message":[],"_revision_number":8},{"id":"38f6604746c8e642ba8f166528a0969b51e53bc3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-09 18:55:09.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"0c10c34121baaaa55a17c0c938c3b8f531c62d91","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-09 19:52:47.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/835881/8/check/tempest-dsvm-full-focal-py3/92c19c9/ : SUCCESS in 2h 06m 27s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/835881/8/check/devstack-plugin-ceph-dsvm-focal-py3/3f14e5d/ : SUCCESS in 2h 21m 00s","accounts_in_message":[],"_revision_number":8},{"id":"af3adac23736788796240aeaa617370e11ef1cee","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-09 19:53:11.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/cf877620876644c9b77970e7c746da5b : SUCCESS in 1h 31m 04s","accounts_in_message":[],"_revision_number":8},{"id":"097c8d7dd93dafa29562978462ca3e0d0a3490eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 19:58:35.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef5a82d72044cb1be626283ec0abdb3 : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7715f3e9963c434196f4c202a6baf539 : SUCCESS in 11m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77aa0be3aca5447e9e7a788768498715 : SUCCESS in 11m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d52ba0ac5064ba7b45bdc12f5e2a277 : FAILURE in 11m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71eef48575264318a58a6d05ea82dddc : SUCCESS in 12m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/02f936967dd740f7976bd16378caeea8 : SUCCESS in 1h 03m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a9710b03a56347b1b75bb5e24ede4733 : SUCCESS in 1h 22m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f28d20e8e0984be6aa7fdf5fe47631ae : SUCCESS in 53m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e11abdd4625c4ff1a3f7dc995219d0d9 : SUCCESS in 15m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6a8466693ed14a0aa54bed28171970eb : SUCCESS in 5m 31s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b5ed8e8f0f0b46ce9fa57f146dcce4bb : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eeafd374924042fa97bc05104d7fa1c4 : SUCCESS in 7m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c1d7b2c4b3594e5caf849114b2dee1c9 : SUCCESS in 6m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/64874670ddcc44ffa35eb37b947eb921 : SUCCESS in 48m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a64be6d276114b67bd4bcc2123834e9c : FAILURE in 6m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fafc02a9f32b44af937b0c0bf13674bc : SUCCESS in 1h 21m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7f9cb4cf25e64274a69fe42293c4d291 : FAILURE in 1h 30m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/00e7c3dd7273490491b81cedc0373145 : SUCCESS in 1h 25m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/be9915fab39e4433a1adad129b7b8ef5 : SUCCESS in 1h 18m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3388bb7846474074bb82a1673e02aab3 : SUCCESS in 1h 29m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/38a3509c9427482bba4e5dd34b26c518 : SUCCESS in 45m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2b48d8d2bb264c4494728f7094588524 : SUCCESS in 1h 06m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7201e7102ef04367ab58b5cb579bede9 : SUCCESS in 2h 27m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0b4ca477c64478198f6e42e81994106 : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":8},{"id":"a952106b4ac78c4a5213144f5f4a5bef27ba46b7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-09 20:21:49.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/8/20220510031341/NEC-V-FC : SUCCESS in 1h 47m 38s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/8/20220510031341/NEC-V-ISCSI : SUCCESS in 2h 08m 05s","accounts_in_message":[],"_revision_number":8},{"id":"d4310a961a7f960b4b436d58df725a986ffd189f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-09 20:25:21.000000000","message":"Patch Set 8:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/09_05_2022_19_58_41\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"629ae003e5efcb463814010f13a323538f4e3665","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-09 20:30:15.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/81/835881/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/bf4b5a9/ : TIMED_OUT in 3h 03m 01s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/81/835881/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f35b4a5/ : SUCCESS in 1h 18m 59s","accounts_in_message":[],"_revision_number":8},{"id":"962f7336c8db4ce36473ab28b00edabe2b5f8def","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-05-09 21:01:04.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c0662e3a0ddb52fcd4e7c2cfe7b13381ff3a35c6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-09 21:11:50.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/81/835881/8/check/DellEMC_PowerFlex/eaf7046/DellEMC_PowerFlex/706 : FAILURE in 1h 12m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"5c41018588507c291a32482a7c61de748f6d93f8","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-05-09 21:25:03.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/8/check/DellEMC_PowerStore_iSCSI/98ae0d6/DellEMC_PowerStore_iSCSI/491 : SUCCESS in 1h 26m 08s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/81/835881/8/check/DellEMC_PowerStore_FC/3ca48eb/DellEMC_PowerStore_FC/516 : SUCCESS in 1h 07m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"ceccd6dad776677421f20cdbfa1039283bd38f41","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-05-09 21:51:04.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/81/835881/8/check/DellEMC_SC_FC/1414c73/DellEMC_SC_FC/578 : FAILURE in 1h 19m 04s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/8/check/DellEMC_SC_iSCSI/a2d5890/DellEMC_SC_iSCSI/599 : SUCCESS in 1h 52m 13s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":8},{"id":"7df7d3c7d1c84b7b660687c50398804cc607847f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-09 22:01:04.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e4cfc3d6f7f5e6dc0b0658a8e3703dccb09344a4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-09 22:01:11.000000000","message":"Patch Set 9:\n\nBuild Started","accounts_in_message":[],"_revision_number":9},{"id":"012eba1e025ce7360a64a947746a79223e63237d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-09 22:01:15.000000000","message":"Patch Set 9:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15968/","accounts_in_message":[],"_revision_number":9},{"id":"e30ab62e4741c3fc58f6ca28d31a198992efccf0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-09 22:02:59.000000000","message":"Patch Set 9:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15968/ : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"bdff4e22a2ad166a4cacf3673601b9d0c6db47de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-09 22:19:41.000000000","message":"Patch Set 9:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1652133671138 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"df616ca671019b4976b0a826f32f7e85ee4830de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 22:21:14.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99d8fa3667a8422b8de9daa2464a48e6 : SUCCESS in 17m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47724e2df1744506b7b820ae85aaa14f : SUCCESS in 18m 08s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"35f98c85ac7c83e89e4259bac55c39cab5518ba0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-09 23:06:11.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/9/check/iscsi-driver-nimble/b04a207/b04a207.html : FAILURE in 1h 00m 16s","accounts_in_message":[],"_revision_number":9},{"id":"c77f97319893a2535676f7288bb45132f6edfe40","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-09 23:20:36.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/835881/9/iscsi/ : SUCCESS in 1h 19m 01s\n- cinder-smb http://cloudbase-ci.com/cinder/835881/9/smb/ : SUCCESS in 1h 14m 35s","accounts_in_message":[],"_revision_number":9},{"id":"0d70ca604d284ef2232aa47c26150c7d2c130e60","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-09 23:20:53.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/835881 : SUCCESS in 1h 18m 22s","accounts_in_message":[],"_revision_number":9},{"id":"e953f7e43adf6d549a3971a72a55536e88107e68","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-09 23:22:03.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/835881 : FAILURE in 1h 19m 33s","accounts_in_message":[],"_revision_number":9},{"id":"0283a3da83a19f2cb9cb2c2078fb8e30c91d8921","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-10 00:07:58.000000000","message":"Patch Set 9:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/835881/9/check/tempest-dsvm-full-focal-py3/995fd28/ : SUCCESS in 1h 43m 10s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/835881/9/check/devstack-plugin-ceph-dsvm-focal-py3/a8944ae/ : SUCCESS in 2h 04m 41s","accounts_in_message":[],"_revision_number":9},{"id":"5f05da4a12442b4fbc2ed0561605d3ec5ec029cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 00:08:13.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a37a1580aebd4b9985e81d9304eedfeb : SUCCESS in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d4397785eb845f7b832fe377dcf26f5 : SUCCESS in 10m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3327cede5a92445b9ecb66620336af03 : SUCCESS in 11m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2775793d22a74751bd4830fc0a228bf2 : SUCCESS in 12m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb88470c974048d396aa1c72b6ef9004 : SUCCESS in 11m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/2ec6e6cef99246f7b7aea1aa4f11dce7 : SUCCESS in 1h 03m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/583471b11e814b4cb19d0783a415a3bb : SUCCESS in 1h 33m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c9dabfc1c034e99b19e188c9830199f : SUCCESS in 56m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c61c4e5b6a2146e1ae4ab9f55d262364 : SUCCESS in 15m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/90dba4d5f9664c94aedd566fa6fbf979 : SUCCESS in 5m 59s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/be8e2f12a0ec477d83a07bffed1740d2 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fcd9f5c41fa240a79ec51be29a10cd5d : SUCCESS in 6m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/76f6614f75d54e06b05ccf1a9f3c87cf : SUCCESS in 6m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/347fceffa8f3485599b7b162e628ec8f : SUCCESS in 52m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a706f132bab64be8b4c8b5b7ffb2e866 : FAILURE in 6m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ead6c26ccd734ecba121655f5178a693 : FAILURE in 1h 05m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dd7e98538fb04d6992872cbaf3341877 : FAILURE in 1h 41m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b5f48ce5767b440cb24ab4e25583d2a8 : SUCCESS in 2h 00m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7cdf03c81f0d40669fc90fdecc13a83f : SUCCESS in 1h 16m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4d4ab20cf189409faeda7055e9d2525e : SUCCESS in 1h 16m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a87972959e50436486b7fbecb6b38316 : SUCCESS in 54m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b40cbf0a4e814b99bb14a751b684f577 : SUCCESS in 1h 23m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/65af49e04f5c4a8f8f1a2affb96488e3 : SUCCESS in 1h 55m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9723700170c6456a84e4f8ff08ed4666 : SUCCESS in 58m 43s","accounts_in_message":[],"_revision_number":9},{"id":"5e0d6e9020997a487742dcaaf515a0ed6822d00b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-10 01:05:39.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/81/835881/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c3c7855/ : TIMED_OUT in 3h 02m 50s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/81/835881/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/68aa2ad/ : SUCCESS in 1h 14m 04s","accounts_in_message":[],"_revision_number":9},{"id":"64d8357486956a2dfda3f6cadda8daa2b4217a0b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-10 02:14:23.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-835881-8/ns5_nfs : FAILURE in 11m 33s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-835881-8/ns5_iscsi : FAILURE in 8h 48m 31s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"bd1fe55e56aa616776c9a35eb2e1d40930648ff4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-10 02:36:48.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"4d2ddfe01cf3d50eff41a56504127209064735a9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-10 03:48:05.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4207005e775a49849e50e8dd7c0647fd : SUCCESS in 1h 27m 34s","accounts_in_message":[],"_revision_number":9},{"id":"22553bd367c1175a4841c521246f22cab6647392","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-05-10 04:13:57.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/8/20220510064814/NEC-FC : SUCCESS in 3h 13m 20s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/8/20220510064814/NEC-ISCSI : SUCCESS in 6h 25m 40s","accounts_in_message":[],"_revision_number":8},{"id":"2549bbcfcb08571673c47ea1ee881f82967184b0","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-10 06:11:37.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/9/20220510124742/NEC-V-FC : SUCCESS in 1h 59m 58s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/9/20220510124742/NEC-V-ISCSI : SUCCESS in 2h 23m 52s","accounts_in_message":[],"_revision_number":9},{"id":"ef20d6d24f167294081818416fdb9b6fb683743c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-10 06:49:56.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-835881-9/ns5_nfs : FAILURE in 8h 47m 22s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-835881-9/ns5_iscsi : FAILURE in 8h 48m 22s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"b555462131c9023f394a8c5a0f46c57c240cbfb6","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-10 13:06:55.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"59e67f1327844d1d80393f97af7b0aad77ab86dc","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-10 13:08:26.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"cc281284b0cbf2d6a191dab59c09f003131c135e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-10 13:11:52.000000000","message":"Patch Set 9: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":9},{"id":"2e0a6adf11480a1469a4b90c4e991e6cb92e6508","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-10 13:17:16.000000000","message":"Patch Set 9:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"a930a0c0d2fcf13dcb35ad7d622239765e7be936","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 13:31:28.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11ffff4ec79544d69560793915e0bd81 : SUCCESS in 20m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2a5641c6bdf4541b6fc421d5f419c8f : SUCCESS in 21m 01s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2900004c8e7e5a79a050ab3b03a10551a80c614a","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-05-10 14:11:12.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"935c67acb785af71729dae79763310d047f39bf4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-10 14:21:53.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/835881/9/iscsi/ : SUCCESS in 1h 13m 00s\n- cinder-smb http://cloudbase-ci.com/cinder/835881/9/smb/ : FAILURE in 57m 51s","accounts_in_message":[],"_revision_number":9},{"id":"0ff3439a467f7100430e10bdcaaa4ee8a8f19779","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 15:32:47.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e0c69e471374718865766f7be8bafcb : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdbbb2f042ef4d28bf4f611d79be6c6b : SUCCESS in 10m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2923105c881a4b33859c2e8706db7946 : SUCCESS in 11m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49b28970241f445b8216a29a10b42ab5 : SUCCESS in 11m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ec5435964fb45a29bff2a24346fe8b0 : SUCCESS in 11m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/d919d948d09e4e66aba12eb476046847 : SUCCESS in 1h 06m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eaa0853423a54cac954838081f779005 : SUCCESS in 1h 20m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c3d7130a8944dfebb6a2f16df91dd9d : SUCCESS in 56m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2bb237a22c5741e88c0cb08fcebe9cc2 : SUCCESS in 15m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b8c7d2ef784b40abac8ff2c71a7deb79 : SUCCESS in 5m 09s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3c218358f1834bc19c4521736f1d8cef : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/23188544c06f4d0dabb67e98cb42e729 : SUCCESS in 5m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ce321f74cd1146b0afe7be745a723f96 : SUCCESS in 6m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1ea24fba142b4e14bd529f4bef7d400b : SUCCESS in 49m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2e900c1eeaaf42518bd5a54289d5aad7 : FAILURE in 6m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e9c25ae5c06b43479bcaf39f9c0bc7ae : SUCCESS in 1h 16m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/37bcc7a15a404a629bcfcde4bb848cd7 : FAILURE in 1h 21m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dc33a2ed264948e5b161327c733cfc4e : SUCCESS in 2h 03m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c32edce2d30346ba885eae7bbdddc2ac : SUCCESS in 1h 04m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9f806d999f3d4e6a91d327689e663566 : SUCCESS in 1h 06m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/92c77bf9aed6416b9e1dcc3f08da6eeb : SUCCESS in 47m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1a57e557a96444909211c0dd5fac18c8 : SUCCESS in 1h 11m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8c405b5f761b42c68d98fbd328949baa : SUCCESS in 2h 17m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb7cc7a262ad4e659b6d4439107afe37 : SUCCESS in 39m 20s","accounts_in_message":[],"_revision_number":9},{"id":"497efeaeae64e7471ba57418702224ad8d970bf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-10 15:55:07.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e8eded9ce32efeeed1a07f7c05cdfd693036427f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-10 15:55:14.000000000","message":"Patch Set 10:\n\nBuild Started","accounts_in_message":[],"_revision_number":10},{"id":"dd68b2ea62335e8a56df16633c62806ba446bd8d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-10 15:55:18.000000000","message":"Patch Set 10:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15971/","accounts_in_message":[],"_revision_number":10},{"id":"5210da5e23fb4617e2f491d62369c62790ce58b6","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-10 15:55:56.000000000","message":"Patch Set 10:\n\n(9 comments)","accounts_in_message":[],"_revision_number":10},{"id":"db1124ea4913602af1339bfac8fd7cb376756720","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-10 15:56:54.000000000","message":"Patch Set 10:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15971/ : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"74ac68a48d1dd355cd45ed5f9bf32c47c6870d50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 16:18:02.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c127b4f683c4d0bbf2ff2ef1c0eeee9 : SUCCESS in 18m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8b21692eee494db286653c065ec93ea7 : SUCCESS in 20m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"974020f6e8cbdead7fddbb8ca61c32f31f9fe533","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-10 16:32:20.000000000","message":"Patch Set 10:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1652198113665 : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"5d5aa8bc9aa59a9d329c5d86b848272d63878908","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-10 17:01:56.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/835881 : FAILURE in 1h 04m 46s","accounts_in_message":[],"_revision_number":10},{"id":"7f576482659b003d2f974cdce08cf0d77b48633e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-10 17:03:47.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/835881/10/iscsi/ : SUCCESS in 1h 08m 08s\n- cinder-smb http://cloudbase-ci.com/cinder/835881/10/smb/ : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":10},{"id":"d72829dd682e70204b71c5e3f7c7a80846eb797d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-10 17:06:51.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/10/check/iscsi-driver-nimble/df0a78f/df0a78f.html : FAILURE in 1h 00m 43s","accounts_in_message":[],"_revision_number":10},{"id":"4541e0c32c500a33e9c3e66e122f9844397e1929","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-05-10 17:37:52.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/9/20220510231444/NEC-ISCSI : SUCCESS in 3h 22m 13s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/9/20220510231444/NEC-FC : SUCCESS in 3h 23m 05s","accounts_in_message":[],"_revision_number":9},{"id":"f8a1acfee34273022bdd8af8d5870f6e5c957111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 18:47:12.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/163656cf84e64de78aa92cab35824208 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1238cf29c8a74fdfa7d8e5a8c421d6e5 : SUCCESS in 10m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7ebbf13b1284b428b3578ed3a2d9c2a : SUCCESS in 10m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0caee3f2114c4b159fd46b794b57a786 : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2da288f92dfc49ac9fd1403b99438af1 : SUCCESS in 12m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/9aa5fd41716e4572b2187d4a96124bc7 : SUCCESS in 1h 05m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ed9ddc7172774116b33e372d799a7431 : SUCCESS in 1h 28m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82203e1cbc9c4b91b8a7cd553a11c5b0 : SUCCESS in 46m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8029a3b804c54b72b2b1fbc6561dc1d4 : SUCCESS in 16m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f001a5389cae4894a899588c311795f3 : FAILURE in 5m 23s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/67f785f51cdf4b2e9594654f8940b88c : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/499860dd451b4429ac9b5b32039c1e02 : SUCCESS in 6m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9510656d8e474baeb5e817c370c0b375 : SUCCESS in 7m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d8af8cf5b5c7431999d0061fe576faa6 : SUCCESS in 46m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4b25e87836bb40f5a24f3c66905b5cc9 : FAILURE in 6m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8e35116c60da46f18f0db189411da711 : SUCCESS in 1h 33m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a3d352503b544df08e9794111b8803b1 : FAILURE in 1h 34m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f976acfd86824ac7b34711ae3cc6265a : SUCCESS in 1h 55m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6da3d94c3b524d20a6aa94efebdae01e : SUCCESS in 1h 13m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f83a3871edb744d58d71cc0ae24570a0 : SUCCESS in 1h 16m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c782d31dfb5d41f4844c5ebda955b97c : SUCCESS in 52m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ad714c4298254094ab8d3b017f303fa0 : SUCCESS in 1h 19m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a6c1ede871414022b5eb2b81fd5f492e : SUCCESS in 2h 45m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a0e3c323fef446384dc4311e3ea44ba : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":10},{"id":"98b5e1edf8e28107548016c7e04f9cb926663ce9","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-10 19:03:40.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"36c234a3e0bf44c861b15738fa748a23f30ce7a8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-05-10 19:11:13.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/10/upstream-check/cinder-cDOT-FCP/40b23ee/ : RETRY_LIMIT in 7m 38s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/10/upstream-check/cinder-cDOT-NFS/7a88024/ : RETRY_LIMIT in 6m 53s\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/10/upstream-check/cinder-cDOT-iSCSI/5711f5b/ : RETRY_LIMIT in 6m 56s","accounts_in_message":[],"_revision_number":10},{"id":"1d3b00dc155c0e70d94a9f341ce534a1ed76ea1c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-10 19:20:44.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/81/835881/10/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/fda344b/ : TIMED_OUT in 3h 03m 32s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/81/835881/10/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/209e403/ : SUCCESS in 1h 10m 13s","accounts_in_message":[],"_revision_number":10},{"id":"0ccd2e00b65d4ff7adb1a2504b10507977bebb6c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-10 19:21:19.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-835881-10/ns5_nfs : SUCCESS in 3h 25m 43s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-835881-10/ns5_iscsi : SUCCESS in 3h 24m 34s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"d5c1953f54a0a350f4ceea3a7624392ba542d47b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-10 19:36:36.000000000","message":"Patch Set 10:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/835881/10/check/tempest-dsvm-full-focal-py3/d8f0032/ : SUCCESS in 2h 14m 57s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/835881/10/check/devstack-plugin-ceph-dsvm-focal-py3/099da3e/ : SUCCESS in 3h 11m 04s","accounts_in_message":[],"_revision_number":10},{"id":"4f02ce9d964969084f24a517375b99cbcadf68d9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-10 19:40:09.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":10},{"id":"42956092b16df3dc59ac55ed8376ee150ab8df1f","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-05-10 19:51:43.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/835881-9/HPMSA-iSCSI/ : SUCCESS in 52:41.62\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/835881-9/HPMSA-FC/ : SUCCESS in 47:46.02","accounts_in_message":[],"_revision_number":9},{"id":"dbac9be26bfe4f16948707e05351106f784a2ac9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-10 20:11:21.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/81/835881/10/check/DellEMC_PowerFlex/63092a6/DellEMC_PowerFlex/711 : FAILURE in 1h 23m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"d211766e3d6093b68241e4b9eeb126b231fa0562","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-05-10 20:11:26.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/10/check/DellEMC_PowerStore_iSCSI/f4f2b55/DellEMC_PowerStore_iSCSI/496 : SUCCESS in 1h 23m 54s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/81/835881/10/check/DellEMC_PowerStore_FC/21eb3b6/DellEMC_PowerStore_FC/521 : SUCCESS in 1h 13m 34s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"5a4122ca5219d3ecffd9844892d241c3f3856df5","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-05-10 20:59:24.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/81/835881/10/check/DellEMC_SC_FC/475ad0d/DellEMC_SC_FC/583 : FAILURE in 1h 20m 19s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/10/check/DellEMC_SC_iSCSI/7387356/DellEMC_SC_iSCSI/604 : SUCCESS in 1h 46m 43s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":10},{"id":"699f22cc080a59a8eba931b019def88c0f4260fc","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-05-10 21:08:19.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/835881-9/PVME-iSCSI/ : SUCCESS in 1:06:24\n- PVME-FC https://os-logs.seagate.com/dell-ci/835881-9/PVME-FC/ : SUCCESS in 1:04:04","accounts_in_message":[],"_revision_number":9},{"id":"e4fbcb712d20eef487a54ac55178db0ad5b963b2","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-10 21:11:17.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/835881-9/Lenovo-iSCSI/ : SUCCESS in 1:14:37\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/835881-9/Lenovo-FC/ : SUCCESS in 1:11:44","accounts_in_message":[],"_revision_number":9},{"id":"006e1788df6f60684cc27ee0e6d6c5aa58f4304d","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-05-10 21:15:20.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/835881-9/Seagate-iSCSI/ : SUCCESS in 1:09:09\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/835881-9/Seagate-FC/ : SUCCESS in 1:06:22","accounts_in_message":[],"_revision_number":9},{"id":"a4e732ee88361f7a5695d71c82943d5fa5931079","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-10 22:06:05.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/10/check/DellEMC_VNX_iSCSI/0ab0315/DellEMC_VNX_iSCSI/371 : SUCCESS in 2h 17m 31s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/81/835881/10/check/DellEMC_VNX_FC/9729cfe/DellEMC_VNX_FC/389 : FAILURE in 14m 28s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":10},{"id":"676a9d2627a4653907a9ee2b9e8fe44cc0a10c93","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-10 23:38:35.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5b616ae6aa054cf1a9f35119bb202507 : SUCCESS in 1h 38m 26s","accounts_in_message":[],"_revision_number":10},{"id":"e50575cca4ca77f2aa229ecf1c4f6e1e068ccc51","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-05-11 00:17:51.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/835881-10/HPMSA-iSCSI/ : SUCCESS in 52:00.85\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/835881-10/HPMSA-FC/ : SUCCESS in 47:02.38","accounts_in_message":[],"_revision_number":10},{"id":"cd6d53baa2a012831ea77bbec1ac8d26bf8373ec","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-05-11 00:30:09.000000000","message":"Patch Set 10:\n\nBuild failed.","accounts_in_message":[],"_revision_number":10},{"id":"4447bbdbdd6d7309e428430e2f640d13d824f237","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-11 01:45:43.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/10/20220511045003/NEC-V-FC : SUCCESS in 1h 59m 52s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/10/20220511045003/NEC-V-ISCSI : SUCCESS in 5h 55m 38s","accounts_in_message":[],"_revision_number":10},{"id":"4a5cd78e4e821e23e33cf5d3e87483ff0a479810","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-05-11 02:30:51.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/835881-10/PVME-iSCSI/ : SUCCESS in 1:03:01\n- PVME-FC https://os-logs.seagate.com/dell-ci/835881-10/PVME-FC/ : SUCCESS in 1:01:46","accounts_in_message":[],"_revision_number":10},{"id":"51d2491402c406df59076e13f8ee8efb57457a97","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-11 02:45:05.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/835881-10/Lenovo-iSCSI/ : SUCCESS in 1:11:38\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/835881-10/Lenovo-FC/ : SUCCESS in 1:10:29","accounts_in_message":[],"_revision_number":10},{"id":"5558719ebde8eb177985a4fb6bd71978e30a185d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-11 13:25:18.000000000","message":"Patch Set 10: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"bde2145a112ba9e7e002cc9e14e2755d8b8fd556","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-11 14:10:51.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"1a19ee4b54a9e1f1b53b59f067913b6fb8a5c27e","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-05-11 14:49:23.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"6b5360acf60aca91d379fa8420cdaa9655bf219d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-05-11 18:07:05.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/10/20220511234715/NEC-ISCSI : SUCCESS in 3h 17m 22s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/10/20220511234715/NEC-FC : SUCCESS in 3h 19m 47s","accounts_in_message":[],"_revision_number":10},{"id":"fc7a749442c53a264a7b6da9fbd8b266947aac3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-12 13:48:08.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"5700c5336baea4ad6219ce7b4fb0921903163c8a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-12 13:51:35.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"52146d2c34797d2955ae3907922d18d5f229ceb1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-12 14:00:56.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":11},{"id":"0030e646020ca700eb26b9b1aeb195652191c656","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 14:31:39.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8cae738ac8446b4a47c175f8838b0d5 : SUCCESS in 20m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cae10a53663a469f948a3507d5bc0615 : SUCCESS in 23m 26s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"636183f75d20dc6983076ff6a77f3a96070115d3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-12 15:09:54.000000000","message":"Patch Set 11:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1652363298754 : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"c227386641487eb45f8231dca36b578d977367e7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-12 15:50:51.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/835881/11/iscsi/ : SUCCESS in 1h 08m 15s\n- cinder-smb http://cloudbase-ci.com/cinder/835881/11/smb/ : SUCCESS in 1h 05m 35s","accounts_in_message":[],"_revision_number":11},{"id":"00ffbf4e6e89ce8d670f57428fd135659c642359","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-12 16:25:03.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/835881 : SUCCESS in 1h 18m 01s","accounts_in_message":[],"_revision_number":11},{"id":"ac5617d5787253aa611e321d62dcb2af132c1124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 16:41:47.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fccd28dce45424c884ae7ada908f721 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da353bbe25ae447d9fde62561bc0c14f : SUCCESS in 11m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b7b71b657d040148823def6a38615ac : SUCCESS in 11m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d417d42832f44cb3b143226be19975e2 : SUCCESS in 12m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0b94ab2990d4c238d878a50e2b8fbf4 : SUCCESS in 13m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/53048df58fbb41c2a3cd41efa986c1d5 : SUCCESS in 1h 04m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f71620ac22aa4e0aa20dfcb46ed6c88c : SUCCESS in 1h 32m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f389821024a7437198db046466feacf8 : SUCCESS in 58m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/df9269de95694ac6a5bb74093c14184e : SUCCESS in 15m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cd8ce9bca18742dcb909256d113227b3 : FAILURE in 5m 37s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/77c9ddd1da4146f1917f4e1782586008 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5c6d5dc6f7ed4cb091f361b9983edd96 : SUCCESS in 7m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6e0d08071f644eb798145547b3889940 : SUCCESS in 6m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/04f5b7f52c3f4ddba5d9bfdbd23a7148 : SUCCESS in 49m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7bf13f1e273e4b05a4d059c5b649215d : FAILURE in 7m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5f8ac743a2764a84a3518322a034be34 : SUCCESS in 1h 19m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ad907d15adb047cbb7a21e21a9c60ac5 : FAILURE in 1h 35m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cf4724415c4c4109839864a133a3b821 : SUCCESS in 2h 01m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/39b473d79cac4817973b795b8fca6944 : SUCCESS in 1h 17m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/203b8bdf6f7f4231817cd2f2aaabc521 : SUCCESS in 1h 34m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a506913d96b04f3b8f5076ce01c5e809 : SUCCESS in 47m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1fa8ed22e5ec462fa51b59169b376a29 : SUCCESS in 1h 50m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9fd068ecb42742898f6b51375f913a30 : SUCCESS in 2h 47m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11b93184f2c842e88b44d2950279cf94 : SUCCESS in 1h 18m 37s","accounts_in_message":[],"_revision_number":11},{"id":"b8b8fb36dabd615de2bc9278d13773ae13ae715f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-12 16:41:54.000000000","message":"Patch Set 11:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":11},{"id":"baac1c0637a0b6cde06260d2a1ba8389e84884e0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-12 17:31:49.000000000","message":"Patch Set 11:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/95a694f612fd49a58a14c8c16e23e694 : POST_FAILURE in 2h 18m 25s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/f0fa32a3e6534f90b857947182719104 : POST_FAILURE in 2h 39m 23s","accounts_in_message":[],"_revision_number":11},{"id":"ef002d36016bbf767424621b992caab0bc8536fb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-05-12 17:35:39.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/11/upstream-check/cinder-cDOT-FCP/2d5a70c/ : RETRY_LIMIT in 7m 23s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/11/upstream-check/cinder-cDOT-NFS/0911249/ : FAILURE in 49m 41s\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/11/upstream-check/cinder-cDOT-iSCSI/5e3fcca/ : FAILURE in 51m 36s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"f94188f3f3f2f15d9256344cee36212cd35be580","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-05-12 17:36:22.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/11/upstream-solidfire-check/cinder-solidfire-iSCSI/eba21b0/ : SUCCESS in 52m 50s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":11},{"id":"88e00f2501d445295870b57eb93b0212d2d45f2b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-12 18:03:20.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/81/835881/11/check/DellEMC_PowerFlex/1ccf3ea/DellEMC_PowerFlex/722 : FAILURE in 1h 21m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"347d8e6d0f5c084217773f69107a67e3cc2c19bb","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-05-12 18:04:39.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/11/check/DellEMC_PowerStore_iSCSI/a51247e/DellEMC_PowerStore_iSCSI/507 : SUCCESS in 1h 22m 31s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/81/835881/11/check/DellEMC_PowerStore_FC/f3eae67/DellEMC_PowerStore_FC/532 : SUCCESS in 1h 12m 01s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":11},{"id":"dc08ecc2579d602a38362fb9f553bfca70e5d81f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-12 18:39:57.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/11/check/iscsi-driver-nimble/8ef6c16/8ef6c16.html : SUCCESS in 1h 36m 30s","accounts_in_message":[],"_revision_number":11},{"id":"cb7a28a3497dbc13565ddf9ab7ba47b3e397f65c","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-05-12 18:49:35.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/835881-11/HPMSA-iSCSI/ : SUCCESS in 49:55.12\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/835881-11/HPMSA-FC/ : SUCCESS in 45:52.85","accounts_in_message":[],"_revision_number":11},{"id":"054528bf76525adbc7a64969d2a03f0c63811978","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-12 19:18:28.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/835881-11/Lenovo-iSCSI/ : SUCCESS in 1:03:57\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/835881-11/Lenovo-FC/ : SUCCESS in 1:00:16","accounts_in_message":[],"_revision_number":11},{"id":"5354602a3486f52ed9458731f6bed4f15182fbae","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-05-12 19:26:48.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/835881-11/PVME-iSCSI/ : SUCCESS in 1:07:56\n- PVME-FC https://os-logs.seagate.com/dell-ci/835881-11/PVME-FC/ : SUCCESS in 1:03:56","accounts_in_message":[],"_revision_number":11},{"id":"d9f6e83d1a58de7501d762eb0f98c2d0e511f51e","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-05-12 19:30:49.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/835881-11/Seagate-iSCSI/ : SUCCESS in 1:10:15\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/835881-11/Seagate-FC/ : SUCCESS in 1:05:05","accounts_in_message":[],"_revision_number":11},{"id":"73639a75fd523fb75d6857f8d2d3f1625ccc80ce","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-12 19:32:28.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-835881-11/ns5_nfs : SUCCESS in 3h 23m 20s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-835881-11/ns5_iscsi : SUCCESS in 4h 25m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"4e5a54c5acfb025bd65b29a7cada925b2c25888c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-12 19:38:27.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/11/check/DellEMC_VNX_iSCSI/0c7b7c5/DellEMC_VNX_iSCSI/382 : SUCCESS in 2h 03m 48s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/81/835881/11/check/DellEMC_VNX_FC/4e28789/DellEMC_VNX_FC/400 : FAILURE in 10m 42s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":11},{"id":"40c7d8bf7ddcb24d52979fed715d6691c0584231","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-12 19:41:08.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"7754192c72b3a38e881137816cc580d148cb1eba","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-12 19:50:04.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c6d3b216b2bf48dcb64ab72fd87b3eb8 : SUCCESS in 1h 27m 46s","accounts_in_message":[],"_revision_number":11},{"id":"ae1e83da7e990a95506bf943a6c337085df107bc","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-13 12:20:14.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/11/20220513151734/NEC-V-FC : SUCCESS in 5h 17m 54s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/11/20220513151734/NEC-V-ISCSI : SUCCESS in 6h 02m 37s","accounts_in_message":[],"_revision_number":11},{"id":"bfe6f7d7974bfc61c24bf321ab9394acecdfa05e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-05-15 04:29:46.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/11/20220515100903/NEC-FC : SUCCESS in 3h 17m 08s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/11/20220515100903/NEC-ISCSI : SUCCESS in 3h 20m 40s","accounts_in_message":[],"_revision_number":11},{"id":"7a5b9d56c29293f2820c6050f1e3fbb7c9f71638","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-16 16:57:23.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"c1a592a9a3ff601ec9dedf2fb60d334ad9bf7f40","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-16 16:57:33.000000000","message":"Patch Set 12:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16016/","accounts_in_message":[],"_revision_number":12},{"id":"48485e87c473144a5b3967af4ebd06da4e7f97dd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-16 16:59:15.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"34a9d54554fdf07f86d532bc5fe522befd69ec4c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-16 16:59:51.000000000","message":"Patch Set 12:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16016/ : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"c58f46e21fcc89f8ec7223d4eee1bfae468e9d6d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-16 17:02:01.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"6c5d75b57e45e4e821be973ab8dd2a830a69bf6f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-16 17:17:11.000000000","message":"Patch Set 12:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1652720253601 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"7042835cc5b8148a9200637b57bbd32aa676743f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 17:23:12.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a30ccc6703e4faaafdfb4cb4f94fdd9 : SUCCESS in 21m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f262c5183584a2298c095515e3b75d4 : FAILURE in 23m 42s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"18ea9d3b685680d349dba72f911e71065f051e5c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-16 18:35:34.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/835881/12/iscsi/ : FAILURE in 1h 37m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/835881/12/smb/ : SUCCESS in 1h 20m 45s","accounts_in_message":[],"_revision_number":12},{"id":"02feef969f367b488a05b2f32856002e937d12b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 18:55:13.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4018eb1b3ab148bc9a41efd99962ce7b : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcf1e1de537641aabcc9e65018d72233 : SUCCESS in 10m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ba07dce002e47ef87f0c4298f10dfd5 : SUCCESS in 16m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3156838c25234216a3a75baa2e92c0a7 : SUCCESS in 12m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52aa685c7ca34c17b4d5ec2cfdbae893 : SUCCESS in 16m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/d05e28a686034bf2a88430468bff0369 : SUCCESS in 45m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f27034c8c5543fcb8744c6cc48c505e : SUCCESS in 1h 06m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8bcc6ba179344259766becb833c33a1 : SUCCESS in 1h 10m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/eb21a687af4b450da20487831412da47 : SUCCESS in 13m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/78d55df2802244f8b5e6d862c020d5bd : SUCCESS in 5m 39s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2b8d0ebf313a48b295567623a94aa610 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a1df968655f6468b92607c3e0ca1f9af : SUCCESS in 6m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f1b7a40f778940a1ba61a6a6bba4e8db : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3268b50708d54f02904e5f71dd44db88 : SUCCESS in 36m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6505038c50e241ac8fa1d484960c52ac : FAILURE in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cbc8b31644394b3b9c95b6b5af248f1c : SUCCESS in 1h 07m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/770eacc5dd544b3fb785e244872d0eb0 : FAILURE in 1h 45m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5d16653e0de741b198128f8cd24b226a : FAILURE in 1h 13m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d68bfff569da4bd7b0f0df3e57b6a6a2 : SUCCESS in 50m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/df66178914e64ca5a0a8c524bf5c0d2e : SUCCESS in 1h 10m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ac5ad462947e44f485d69f617ed435f3 : SUCCESS in 36m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/85e5e3022885420dbc949e7dbd376f0d : SUCCESS in 1h 05m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/52aa800c39a64508af2a7df8a0260ff9 : SUCCESS in 1h 53m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a0aa1139a7744abbe1698a3e905cc4f : SUCCESS in 46m 32s","accounts_in_message":[],"_revision_number":12},{"id":"39680af7057c45d2942aefb6990f41600b81fc7e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-16 19:35:11.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/835881 : FAILURE in 1h 19m 23s","accounts_in_message":[],"_revision_number":12},{"id":"b57a921e46e383ce83feb7c979746371d9f3fa9e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-16 20:24:57.000000000","message":"Patch Set 12:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/f91abf9fa59a44d597728e82f3f37159 : POST_FAILURE in 2h 33m 52s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/61e7bcb2fe1346cf98ccfa78e3f73cc0 : TIMED_OUT in 3h 09m 40s","accounts_in_message":[],"_revision_number":12},{"id":"bc3a949e49721135fcb66703d29aa0eb5707a1d4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-16 20:50:55.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-835881-12/ns5_nfs : SUCCESS in 3h 44m 18s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-835881-12/ns5_iscsi : SUCCESS in 3h 53m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"7890ad70ce056ae751c2de0da95ea13f8df4619b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-16 21:32:58.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/81/835881/12/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4565585/ : TIMED_OUT in 3h 02m 56s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/81/835881/12/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/558596f/ : SUCCESS in 1h 18m 46s","accounts_in_message":[],"_revision_number":12},{"id":"c1cdd4f8423d5efd38231df0e924e3c682528916","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-16 23:28:20.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/12/check/iscsi-driver-nimble/102fba7/102fba7.html : SUCCESS in 1h 39m 16s","accounts_in_message":[],"_revision_number":12},{"id":"de939aa14e1df5556ae7069717c299259a4c4ed5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-17 00:28:32.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/81/835881/12/check/DellEMC_Unity_FC/9596458/DellEMC_Unity_FC/756 : SUCCESS in 1h 05m 34s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/12/check/DellEMC_Unity_iSCSI/a2034a5/DellEMC_Unity_iSCSI/726 : SUCCESS in 1h 42m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"139c1f9343552ee7a2840c745ea93263622c3d7a","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-17 13:40:25.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"cc25b415c631aa79876dca5778664b7d6b12ab2c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-17 13:42:15.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"e471caf562b6f04137708c7c10cb935ad5c08e31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 14:14:47.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f9ef535993842b081c0e5e595674a84 : FAILURE in 29m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3ac8216058942fd80550d454b7e6fff : FAILURE in 32m 37s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d291ebd22805f27fe363cf5fd903cd960f597217","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-17 14:55:24.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/835881/12/iscsi/ : SUCCESS in 1h 14m 22s\n- cinder-smb http://cloudbase-ci.com/cinder/835881/12/smb/ : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":12},{"id":"413e83a7052e53821b439070fb8eea9dbebce74e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-17 14:58:14.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"ea94f53dc4d2fccb8e7d3f8508ae91f7042c4d59","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-05-17 15:04:30.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/37f96ea/index.html : SUCCESS in 1h 20m 29s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/0872fcd/index.html : SUCCESS in 1h 23m 23s","accounts_in_message":[],"_revision_number":12},{"id":"dbbfc523205802b9e0f3522ef1d49ac249386295","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 16:11:05.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d7b473acf845e9a375fdb02371d843 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a5277202f824b1ab674b40e14e8faf1 : SUCCESS in 10m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a7b510175ca475fb5bf8bdc5066d293 : SUCCESS in 11m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8712cfba19a04d7a83b0755c1c9c222a : SUCCESS in 14m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7efb24066bfe4f80a9f2d0e9122fe9e5 : SUCCESS in 10m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/500a840d0de940aa99520556e2c35edf : SUCCESS in 1h 05m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/68b64b2ddaaf463f8cb7d7c6a5cb39dc : SUCCESS in 1h 23m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96291607b3e84d898a568b1e615e65a6 : SUCCESS in 52m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c3de212708fc49a39a59f40c4b450a26 : SUCCESS in 16m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f41b8abe282e4c40a8003bdbf175ecaf : SUCCESS in 6m 00s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0d99a8fa36654438b96a8ac0bc1c493e : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d21b483b30824c46b60a100ccc992d6a : SUCCESS in 5m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/477e86b64a9349a0aef122ba2871591e : SUCCESS in 7m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e44c30d4ff23416ebaefaba9d0da69df : SUCCESS in 43m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/75adb1540c504492881e089e9f14f852 : FAILURE in 6m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/04e1b06655d548ae9a846b1a88b49b23 : SUCCESS in 58m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f1ea90578d6a4bd0b6ce828f78207f5b : FAILURE in 1h 22m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cca3dcc6448c43b4a181961ecd2ccd97 : SUCCESS in 1h 43m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/34fd2f8014b94a7f8ddfb82eb5760220 : SUCCESS in 57m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5ebb79ccba46422992cfa0978b0bb3aa : SUCCESS in 1h 08m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6e6f046942e94462a27f02d5d0199ce3 : SUCCESS in 42m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/70b2e45d9d6d4901a75a7197e5c713c2 : SUCCESS in 1h 21m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/91432b8621c34d8592df08a5a755b7e8 : SUCCESS in 2h 25m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4063da102f447718369fdbac45aa53e : SUCCESS in 56m 14s","accounts_in_message":[],"_revision_number":12},{"id":"320644bf2564a82e020dbc8813cc74766227214c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-17 16:11:12.000000000","message":"Patch Set 12:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":12},{"id":"632889ee23506f9ce520378b2e234880513184f3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-17 16:31:42.000000000","message":"Patch Set 12:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/a0e9fd8aa30c4ff7a77375fecdecb903 : POST_FAILURE in 2h 47m 44s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/07b7804cd4df48bca97f17f71adf2242 : POST_FAILURE in 2h 37m 37s","accounts_in_message":[],"_revision_number":12},{"id":"350af79a7849a876ecf49edee57e2b77cdc38d86","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-05-17 17:05:29.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/12/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/1067657/ : SUCCESS in 52m 58s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":12},{"id":"d99370a4459ed45b07fc07438e1b982cb865a406","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-05-17 17:18:21.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/12/upstream-check/cinder-tempest-plugin-ontap-fcp/7653ce2/ : SUCCESS in 1h 01m 15s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/12/upstream-check/cinder-tempest-plugin-ontap-iscsi/fee4eb8/ : FAILURE in 50m 58s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/12/upstream-check/cinder-tempest-plugin-ontap-nfs/2a464ff/ : FAILURE in 47m 15s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"1e870f3803a635e0cf27bced4e7863835d300ad3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-17 17:23:38.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/81/835881/12/check/DellEMC_PowerFlex/687c012/DellEMC_PowerFlex/747 : SUCCESS in 1h 12m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"bc03b2ded3caef4f54153e0ecabd5a5c48af5192","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-05-17 17:43:18.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/81/835881/12/check/DellEMC_SC_FC/5f58aa8/DellEMC_SC_FC/593 : SUCCESS in 42m 30s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/12/check/DellEMC_SC_iSCSI/d0fe5cf/DellEMC_SC_iSCSI/614 : SUCCESS in 1h 31m 55s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":12},{"id":"a58a4cf1c7dc9025c92040f9d3d0740f023defc1","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-17 18:15:15.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/12/check/DellEMC_VNX_iSCSI/fb11372/DellEMC_VNX_iSCSI/403 : FAILURE in 2h 03m 51s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/81/835881/12/check/DellEMC_VNX_FC/ca06e50/DellEMC_VNX_FC/422 : FAILURE in 13m 05s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":12},{"id":"52bd648d0eaee36ba8f76840e67686f9edfd0c97","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-05-17 18:38:19.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/835881-12/HPMSA-iSCSI/ : SUCCESS in 55:31.58\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/835881-12/HPMSA-FC/ : SUCCESS in 56:57.34","accounts_in_message":[],"_revision_number":12},{"id":"20ddc4d73e0f5667de2bf3b821ddfab9d4f8612b","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-17 20:42:25.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/835881-12/Lenovo-iSCSI/ : SUCCESS in 1:05:45\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/835881-12/Lenovo-FC/ : SUCCESS in 1:03:30","accounts_in_message":[],"_revision_number":12},{"id":"1944368490d65aa4612c61d2be47ea4e25829905","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-05-17 20:43:10.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/835881-12/PVME-iSCSI/ : SUCCESS in 1:06:32\n- PVME-FC https://os-logs.seagate.com/dell-ci/835881-12/PVME-FC/ : SUCCESS in 1:04:05","accounts_in_message":[],"_revision_number":12},{"id":"58fb8f53697353a4675420e499d9c6432072cec7","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-05-17 21:12:10.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/835881-12/Seagate-iSCSI/ : SUCCESS in 1:11:18\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/835881-12/Seagate-FC/ : SUCCESS in 1:04:26","accounts_in_message":[],"_revision_number":12},{"id":"d5cb09c3c54f28f26c96d8b6081233d4656cba23","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-17 22:28:36.000000000","message":"Patch Set 12: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"45d50f0f86173e8484aedc6bc8d5e5d3e315b89e","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-05-17 23:42:17.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"9112f753865e79fe8b09ac08ffcfd91a8c975776","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-18 11:46:14.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/81/835881/12/check/DellEMC_Unity_FC/d0e73f0/DellEMC_Unity_FC/776 : SUCCESS in 1h 16m 33s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/12/check/DellEMC_Unity_iSCSI/49fe801/DellEMC_Unity_iSCSI/745 : SUCCESS in 1h 33m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"bc29ea7a6780c1bc549b0195f863aa91a1dc5d64","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-05-18 17:17:05.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/12/20220518195731/NEC-ISCSI : SUCCESS in 3h 20m 12s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/12/20220518195731/NEC-FC : SUCCESS in 6h 19m 31s","accounts_in_message":[],"_revision_number":12},{"id":"9af8bafe2b4223fce1ff6ec6e0d5b134b099b058","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-18 20:52:46.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/12/20220519034659/NEC-V-FC : SUCCESS in 1h 46m 35s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/12/20220519034659/NEC-V-ISCSI : SUCCESS in 2h 05m 44s","accounts_in_message":[],"_revision_number":12},{"id":"315cf1e0b3550187a69c3bb17a066f79bb5a01d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-19 18:34:02.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"4f1956a8fbbaf94efccc4025785271121ff9ab98","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-19 18:34:15.000000000","message":"Patch Set 13:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16046/","accounts_in_message":[],"_revision_number":13},{"id":"f2aa62def67a3c3488231f949b058d983a25a719","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-19 18:36:23.000000000","message":"Patch Set 13:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16046/ : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"3c8fc85e4bbf846671a6cf783763fbd7b5b827a5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-19 18:36:33.000000000","message":"Patch Set 13:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":13},{"id":"07e7e1b28a3d264394f5e7cb3f12c0a537daa864","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-19 18:45:49.000000000","message":"Patch Set 13:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-835881-13/ns5_nfs : FAILURE in 11m 15s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-835881-13/ns5_iscsi : FAILURE in 11m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"096dff5e37f4dcdcc7af7fca6cf01fb4b80e9eec","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-19 18:58:18.000000000","message":"Patch Set 13:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":13},{"id":"fcea17d35511e0b28e4676fcfef7c1b13dfc8581","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 19:01:48.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/15a021965ca64fb0a5fb4c06f7265647 : FAILURE in 11m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3ead30a142794c9cb1484a5913e89f09 : SUCCESS in 25m 14s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"7299cec0dcbab4a9ed08fc9f046ee8de961fbc9d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-19 19:02:48.000000000","message":"Patch Set 13:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1652985251669 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"6b7c8f13214a6c6cdd1f2278959cfeb1e6a89d29","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-19 19:07:36.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/81/835881/13/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4e8d223/ : FAILURE in 29m 29s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/81/835881/13/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ed89f98/ : FAILURE in 31m 21s","accounts_in_message":[],"_revision_number":13},{"id":"f66685220123db8cd20025d54e2e958fc64b71cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 19:32:30.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46dda60c9cc149d5bfc788f43137043f : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/180f0e8ab0b7448e9466b70f69ae4c21 : FAILURE in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15c9d275441d4098872d9a5b800b30a7 : SUCCESS in 11m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3201175a36de4877b66c5396cb57f8af : SUCCESS in 13m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d535e7168b4986835973531db7fca2 : FAILURE in 8m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/fb78f45be38340df8f5025ae1f60ef06 : FAILURE in 50m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fd507a07215a4f9eb890efa3e04c34bb : FAILURE in 25m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a21ef2e75774f4385325a729caa2ca0 : POST_FAILURE in 28m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/17db5315391e4f99870590fbf5ee8a88 : FAILURE in 5m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4a45ac2710af461db53d92a5257ee758 : FAILURE in 6m 19s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b9e1edd2695e44b188302f67ed42ccf4 : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b9ceb4b08000430d921450cb5627ab0e : FAILURE in 6m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a899b76de6224e08a6b0343124267f76 : SUCCESS in 9m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8592eb7583c944c4a5e5f53dcab8edc1 : FAILURE in 28m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5ce847a107304058a5e31e7a7f0ac838 : FAILURE in 7m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/627e6554ab6a4b0798c0704162394457 : FAILURE in 29m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/103151affee4411b8d894ca5c331a454 : FAILURE in 33m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/968919e0521140648d983d733cdab7fd : FAILURE in 28m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/eedce546f1ac4b5c8059bf6fc4d615fc : FAILURE in 54m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1a54453c8f1c407ca27c42a875012fee : FAILURE in 27m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6944911c8a2d44b28e2ed38d65cf1782 : FAILURE in 25m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/996aeab35d1e4a30bf4ce5fd5e1e04b7 : FAILURE in 27m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0bfa518da87242c3a555e99dd90d9a44 : FAILURE in 30m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/945e4f2dd2e64975adde4e31991a9e9b : FAILURE in 25m 45s","accounts_in_message":[],"_revision_number":13},{"id":"4d1996c45d0646e2bf7c0a4ff6d428d413baa0a5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-19 19:35:59.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/835881 : FAILURE in 59m 25s","accounts_in_message":[],"_revision_number":13},{"id":"e252fe819f2ab671458cbb7edec51a39871b61ac","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-19 20:36:34.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/835881/13/iscsi/ : FAILURE in 1h 54m 58s\n- cinder-smb http://cloudbase-ci.com/cinder/835881/13/smb/ : FAILURE in 2h 01m 53s","accounts_in_message":[],"_revision_number":13},{"id":"2a37b45be98422f316861b010f2ad99077781505","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-19 22:55:47.000000000","message":"Patch Set 13:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/84f9599daab6452995115188cb145187 : POST_FAILURE in 1h 38m 00s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/533d29c884a2448eadc350d33c88816d : POST_FAILURE in 3h 02m 16s","accounts_in_message":[],"_revision_number":13},{"id":"045636afee71af1b44bfe64d43b6b3cd29075415","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-20 01:19:59.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/13/check/iscsi-driver-nimble/2163721/2163721.html : FAILURE in 58m 14s","accounts_in_message":[],"_revision_number":13},{"id":"e7f1c546e9b293ec3e804d88e9c37e89a784e766","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-20 03:11:20.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/81/835881/13/check/DellEMC_Unity_FC/bed36e4/DellEMC_Unity_FC/801 : FAILURE in 43m 16s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/13/check/DellEMC_Unity_iSCSI/48ba654/DellEMC_Unity_iSCSI/769 : FAILURE in 35m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"fdd3ea9b94370b27b05309fae458a916f7cf4306","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-20 18:32:33.000000000","message":"Patch Set 13:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-835881-13/ns5_nfs : FAILURE in 32m 32s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-835881-13/ns5_iscsi : FAILURE in 27m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"80ebc8372d9f8357a74b80a99288be2a24639ac4","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-22 09:58:57.000000000","message":"Patch Set 13:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/13/20220522180341/NEC-V-ISCSI : FAILURE in 51m 15s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/13/20220522180341/NEC-V-FC : FAILURE in 55m 13s","accounts_in_message":[],"_revision_number":13},{"id":"d83261c3509110dc659cea10726217e9b39bb0e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-23 14:36:24.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"2172af9e671bdb0f69d3270a498733bf3f3b26cf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-23 14:36:37.000000000","message":"Patch Set 14:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16057/","accounts_in_message":[],"_revision_number":14},{"id":"a46c5dcc607508f139ba2aecdd443b847f687af1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-23 14:38:42.000000000","message":"Patch Set 14:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16057/ : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"c8b99558339717eea7a19db9f127302598816c5e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-23 14:40:42.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":14},{"id":"4eb0687bdf3fbdfcb54ab4310bf737b45d36b344","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-23 15:02:43.000000000","message":"Patch Set 14:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1653316590882 : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"0f022c465dd9747edb43dd7c629e4610bfdb7e61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 15:21:49.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f9d8cd118174e6897490edf73449c37 : FAILURE in 29m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9c642471f377464bbfb03c90aa028976 : SUCCESS in 20m 43s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"826ae8cd64a2acfd2db541137f27ee6810883880","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-23 16:30:18.000000000","message":"Patch Set 14:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/835881/14/iscsi/ : FAILURE in 1h 53m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/835881/14/smb/ : SUCCESS in 1h 06m 40s","accounts_in_message":[],"_revision_number":14},{"id":"e1d9a51d1f9610d18b7745ef6927c2f6420f449c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-23 16:45:10.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d2582167567e49c283685adf7eb87bcb : RETRY_LIMIT in 11m 20s","accounts_in_message":[],"_revision_number":14},{"id":"c34dc209d69c3e2f15ba796673b8227c1fd5fb76","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-23 16:49:10.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/835881/14/check/iscsi-driver-nimble/74deee8/74deee8.html : SUCCESS in 1h 37m 00s","accounts_in_message":[],"_revision_number":14},{"id":"f8c7978286e03c3c076054b555d70a9749893256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 17:22:53.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71204e66525d4b639c1b989c31a8492a : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fae9b25bbeb4e309407c1a64376854c : SUCCESS in 11m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdb1effe318b4a5384612e30751cb892 : SUCCESS in 13m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ac964c25c994d1b97f027beab177558 : SUCCESS in 11m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb30ab92cbc24065bfb2302a6ead835a : SUCCESS in 12m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/ee5fbdbf703c4464a988a26132179675 : SUCCESS in 58m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ee3d7d191e704e73831b28e3330030f3 : SUCCESS in 1h 16m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c007ff733b934c3ab4e2c224fdfc6ffe : SUCCESS in 1h 08m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ee34ce780e36481ebc1aa509fc7210e5 : SUCCESS in 14m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e9fbd0280b6740eead1bf53ea0616c89 : SUCCESS in 5m 57s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d0a1b5957d9a429cabadc1ee8c81087d : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/81ec008ae7f54270b01fb8e6e829f805 : SUCCESS in 6m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a95fd6095422481eb74772dbbcf4e8ac : SUCCESS in 6m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bb093ae82e4c4329856bb067f247a40b : SUCCESS in 44m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/121dc12fd0264c258581a20629a68c08 : FAILURE in 6m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8d64d027fc8e4760bd2775f2de27358d : SUCCESS in 1h 28m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3501d129e5ac417fb2c9b87b4cc7a408 : FAILURE in 1h 25m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9a9d8ae3a036488aa05a571feeb661d8 : SUCCESS in 1h 59m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b47c79a3848c4fa38e93225194f48011 : SUCCESS in 1h 10m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1449c642ced14842a25e42b8d854f930 : SUCCESS in 1h 25m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/851245cc4ad44c819be77bd17781b213 : SUCCESS in 50m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/aff2104af3ef4390a583d57351a5c045 : SUCCESS in 1h 27m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4baad1d311a5466b97090ba58d927e8b : SUCCESS in 2h 40m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d6d648e465841bea65483907517dde5 : SUCCESS in 54m 31s","accounts_in_message":[],"_revision_number":14},{"id":"4c953bde51ece1979a2375bd433a80ca4bb7c587","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-23 17:23:03.000000000","message":"Patch Set 14:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":14},{"id":"269eb77d60d1c938cd219776f5fbe7872a22e246","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-23 17:41:10.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/14/check/DellEMC_VNX_iSCSI/c545aed/DellEMC_VNX_iSCSI/438 : FAILURE in 1s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/81/835881/14/check/DellEMC_VNX_FC/ffd02df/DellEMC_VNX_FC/457 : FAILURE in 9m 47s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":14},{"id":"25c8a0d232ce60e3c87fec2803fcdfa91862ac29","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-23 17:41:47.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/81/835881/14/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4812b07/ : TIMED_OUT in 3h 03m 27s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/81/835881/14/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d2f336c/ : FAILURE in 1h 28m 06s","accounts_in_message":[],"_revision_number":14},{"id":"760616e05cda6b3a3a830f91640e00068d79ab94","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-23 18:14:49.000000000","message":"Patch Set 14:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-835881-14/ns5_nfs : SUCCESS in 3h 37m 53s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-835881-14/ns5_iscsi : SUCCESS in 3h 37m 20s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":14},{"id":"7c817a187bbb2926aeca4051cc09b1daa932ce92","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-05-23 18:33:44.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"611c7e0687f5ed7981ae7dd438320034259a7530","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-23 18:41:49.000000000","message":"Patch Set 14:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/81/835881/14/check/devstack-plugin-ceph-dsvm-focal-py3/f0a2385/ : SUCCESS in 3h 01m 05s","accounts_in_message":[],"_revision_number":14},{"id":"d57b182f6e51b6599113d1819365a7548e39594d","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-05-23 19:11:23.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/14/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/7975c9f/ : SUCCESS in 50m 57s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":14},{"id":"666f9819991ed7bb99f5cc425554fc88f0e03f04","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-05-23 19:12:43.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/14/upstream-check/cinder-tempest-plugin-ontap-fcp/689815b/ : SUCCESS in 1h 31m 57s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/14/upstream-check/cinder-tempest-plugin-ontap-iscsi/1933347/ : FAILURE in 1h 47m 23s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/81/835881/14/upstream-check/cinder-tempest-plugin-ontap-nfs/9825346/ : FAILURE in 1h 42m 40s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"22db801c7f9e17e082b92cc49aeb9a45a49ed125","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-23 20:05:27.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/14/20220524025221/NEC-V-FC : SUCCESS in 1h 58m 44s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/14/20220524025221/NEC-V-ISCSI : SUCCESS in 2h 13m 03s","accounts_in_message":[],"_revision_number":14},{"id":"d31247e6ed52428d213c2de36c71e701568abe9f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-23 21:44:57.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/81/835881/14/check/DellEMC_PowerFlex/a2a3d61/DellEMC_PowerFlex/781 : SUCCESS in 1h 09m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":14},{"id":"ea0dd741220fa68178c0f19443e17b068dba7b93","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-05-23 23:31:24.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/14/check/DellEMC_PowerStore_iSCSI/1524c2a/DellEMC_PowerStore_iSCSI/526 : SUCCESS in 1h 20m 23s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/81/835881/14/check/DellEMC_PowerStore_FC/9b9b6a9/DellEMC_PowerStore_FC/551 : SUCCESS in 1h 11m 08s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":14},{"id":"6cbfc00bbc72cedae4a348389c11698f39957382","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-05-24 01:39:40.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/81/835881/14/check/DellEMC_SC_FC/3da5118/DellEMC_SC_FC/628 : SUCCESS in 50m 25s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/14/check/DellEMC_SC_iSCSI/2049f7f/DellEMC_SC_iSCSI/649 : SUCCESS in 1h 33m 29s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":14},{"id":"1fd7012ef5411fcf93833fd681920bed082dade9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-24 01:40:08.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/81/835881/14/check/DellEMC_Unity_FC/f05e5bd/DellEMC_Unity_FC/816 : SUCCESS in 1h 12m 23s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/81/835881/14/check/DellEMC_Unity_iSCSI/a00b38f/DellEMC_Unity_iSCSI/784 : SUCCESS in 1h 40m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"da6478ff04f7838c803f719a7bb58ad6a60fad0e","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-05-24 02:39:29.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/835881-14/HPMSA-iSCSI/ : SUCCESS in 52:12.66\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/835881-14/HPMSA-FC/ : SUCCESS in 49:39.77","accounts_in_message":[],"_revision_number":14},{"id":"0993933f615175f8ac8315075a321f04d303af05","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-05-24 06:34:11.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/835881-14/PVME-iSCSI/ : SUCCESS in 1:03:38\n- PVME-FC https://os-logs.seagate.com/dell-ci/835881-14/PVME-FC/ : SUCCESS in 1:01:43","accounts_in_message":[],"_revision_number":14},{"id":"119cec5d2af50dfe0cdec0688f56726e9b83a5da","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-05-24 07:48:56.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/835881-14/Seagate-iSCSI/ : SUCCESS in 1:05:46\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/835881-14/Seagate-FC/ : SUCCESS in 1:06:21","accounts_in_message":[],"_revision_number":14},{"id":"b5080367119efd0ddfc8f25807cd29a33655159b","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-24 13:54:40.000000000","message":"Patch Set 14:\n\n(4 comments)","accounts_in_message":[],"_revision_number":14},{"id":"017ca34105b952b3c226a59c8d5e6bedc7daf11b","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-24 13:54:51.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"8eb82639db191c2cb25ede1e511ae9a6e01f4a72","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-05-24 13:56:55.000000000","message":"Patch Set 14: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"3aa4b43606403d73c805bbe4455ed05bf8c4d6f5","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-05-24 14:28:08.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"43e5586f14697974b6a6dec5779abe3e54c616fe","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-25 03:09:32.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/835881-14/Lenovo-iSCSI/ : SUCCESS in 1:04:09\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/835881-14/Lenovo-FC/ : SUCCESS in 1:03:24","accounts_in_message":[],"_revision_number":14},{"id":"910e95d6c7512f6c37ff9736b56b6c2d1a562195","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-05-25 12:30:51.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/14/20220525180852/NEC-FC : SUCCESS in 3h 19m 45s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/81/835881/14/20220525180852/NEC-ISCSI : SUCCESS in 3h 21m 55s","accounts_in_message":[],"_revision_number":14},{"id":"d05f820bf8cf9b2b9693b98df78d4bf46475de6f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-27 15:51:29.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d9dd77547474f8c8a2a781da5401f0626981f8ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 15:53:02.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"ce58ef4b2210b448c1ecc31394484b21f35940aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 18:32:29.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe51a39c1bf5412f9ae56b96ec680bd0 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebaab487e218413798632d8002d9e0bb : SUCCESS in 11m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f01715a673ba42f2b4955de006256508 : SUCCESS in 12m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab5ff534b6ee4cf2aa629b1d0141ac88 : SUCCESS in 12m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/606d1134224c4b128b30342e08200111 : SUCCESS in 53m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c7ad332b09f4fde984b9bf28f9f8d07 : SUCCESS in 1h 18m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8aa57fc7509540e9bc9af217729dab39 : SUCCESS in 47m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2ba0e6e1e3014938a3ad1e3684ad84ad : SUCCESS in 1h 02m 56s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/191c785610a9424fb5905e71ad96c271 : SUCCESS in 1h 18m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d7b9fe8a58148ee9261068dc7069515 : SUCCESS in 47m 19s","accounts_in_message":[],"_revision_number":14},{"id":"fbc99e2ff021b916003a75f05cfbd6b76542ebb2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 18:32:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"adc417ca5304a205d3580c40889eafd3d5cb5f3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 18:38:54.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4957d05a1b89463ca4a56ed4d95a0e14 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"04b6700f48f0776f7bedbea7a26ff18070117c1e","revisions":{"955da9b2f82a95ab82aaea77b04337cf9b0d717a":{"kind":"REWORK","_number":1,"created":"2022-03-30 17:25:06.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/81/835881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/835881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/835881/1"}}},"commit":{"parents":[{"commit":"8b277f717a67329d67c68b773d03f188f467d58b","subject":"WIP: mypy: objects/backup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b277f717a67329d67c68b773d03f188f467d58b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-03-30 16:36:21.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-03-30 17:24:00.000000000","tz":-240},"subject":"mypy: cinder/volume/volume_types.py","message":"mypy: cinder/volume/volume_types.py\n\nChange-Id: I89fc0f75d4839fedc4fc771e6d23b9526109cf4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/955da9b2f82a95ab82aaea77b04337cf9b0d717a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/955da9b2f82a95ab82aaea77b04337cf9b0d717a"}]},"branch":"refs/heads/master"},"f1f961f6ceffb0043082c1df3397b923f943d55b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-31 21:54:46.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/81/835881/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/835881/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/835881/2"}}},"commit":{"parents":[{"commit":"861d83e3341939e36b6bef9127bd8a56bd99b506","subject":"WIP: mypy: objects/backup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/861d83e3341939e36b6bef9127bd8a56bd99b506"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-03-30 16:36:21.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-03-31 21:54:46.000000000","tz":0},"subject":"mypy: cinder/volume/volume_types.py","message":"mypy: cinder/volume/volume_types.py\n\nChange-Id: I89fc0f75d4839fedc4fc771e6d23b9526109cf4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f1f961f6ceffb0043082c1df3397b923f943d55b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f1f961f6ceffb0043082c1df3397b923f943d55b"}]},"branch":"refs/heads/master","description":"Rebase"},"2d49185a7b988de5b62db477d961f819c2886b34":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-04-04 14:49:50.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/81/835881/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/835881/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/835881/3"}}},"commit":{"parents":[{"commit":"5979de2c6189ac075b9da7e245c9645009475c77","subject":"WIP: mypy: objects/backup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5979de2c6189ac075b9da7e245c9645009475c77"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-03-30 16:36:21.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-04-04 14:49:50.000000000","tz":0},"subject":"mypy: cinder/volume/volume_types.py","message":"mypy: cinder/volume/volume_types.py\n\nChange-Id: I89fc0f75d4839fedc4fc771e6d23b9526109cf4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2d49185a7b988de5b62db477d961f819c2886b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2d49185a7b988de5b62db477d961f819c2886b34"}]},"branch":"refs/heads/master","description":"Rebase"},"64712d8ed694a6be7135f753bebbe13056f0ae3e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-04-05 21:47:05.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/81/835881/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/835881/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/835881/4"}}},"commit":{"parents":[{"commit":"30ab8d1cb23ddb94797ce37be590ad6e07d0504b","subject":"WIP: mypy: objects/backup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/30ab8d1cb23ddb94797ce37be590ad6e07d0504b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-03-30 16:36:21.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-04-05 21:47:05.000000000","tz":0},"subject":"mypy: cinder/volume/volume_types.py","message":"mypy: cinder/volume/volume_types.py\n\nChange-Id: I89fc0f75d4839fedc4fc771e6d23b9526109cf4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64712d8ed694a6be7135f753bebbe13056f0ae3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64712d8ed694a6be7135f753bebbe13056f0ae3e"}]},"branch":"refs/heads/master","description":"Rebase"},"6fc5889ff6dc5d527253769c3c5448f758f96ccb":{"kind":"REWORK","_number":5,"created":"2022-04-06 13:08:43.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/81/835881/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/835881/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/835881/5"}}},"commit":{"parents":[{"commit":"30ab8d1cb23ddb94797ce37be590ad6e07d0504b","subject":"WIP: mypy: objects/backup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/30ab8d1cb23ddb94797ce37be590ad6e07d0504b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-03-30 16:36:21.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-04-06 12:58:52.000000000","tz":-240},"subject":"mypy: cinder/volume/volume_types.py","message":"mypy: cinder/volume/volume_types.py\n\nChange-Id: I89fc0f75d4839fedc4fc771e6d23b9526109cf4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6fc5889ff6dc5d527253769c3c5448f758f96ccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6fc5889ff6dc5d527253769c3c5448f758f96ccb"}]},"branch":"refs/heads/master"},"3e561e4c46967bd727a2c4e51fb58cc60a15c74b":{"kind":"REWORK","_number":6,"created":"2022-05-05 14:30:45.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/81/835881/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/835881/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/835881/6"}}},"commit":{"parents":[{"commit":"4f73ff92091666a91c97be8269939c2159edd50e","subject":"Merge \"mypy: policy.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4f73ff92091666a91c97be8269939c2159edd50e"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-03-30 16:36:21.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-05 14:29:12.000000000","tz":-240},"subject":"mypy: cinder/volume/volume_types.py","message":"mypy: cinder/volume/volume_types.py\n\nChange-Id: I89fc0f75d4839fedc4fc771e6d23b9526109cf4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e561e4c46967bd727a2c4e51fb58cc60a15c74b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e561e4c46967bd727a2c4e51fb58cc60a15c74b"}]},"branch":"refs/heads/master"},"d680a678da31a0a325b80d136b0931557caf2d70":{"kind":"REWORK","_number":7,"created":"2022-05-09 14:22:20.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/81/835881/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/835881/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/835881/7"}}},"commit":{"parents":[{"commit":"4f73ff92091666a91c97be8269939c2159edd50e","subject":"Merge \"mypy: policy.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4f73ff92091666a91c97be8269939c2159edd50e"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-03-30 16:36:21.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-09 14:21:30.000000000","tz":-240},"subject":"mypy: cinder/volume/volume_types.py","message":"mypy: cinder/volume/volume_types.py\n\nChange-Id: I89fc0f75d4839fedc4fc771e6d23b9526109cf4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d680a678da31a0a325b80d136b0931557caf2d70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d680a678da31a0a325b80d136b0931557caf2d70"}]},"branch":"refs/heads/master"},"187142ccaef8e942ad9b34baa4e8ab9b8f42af9e":{"kind":"REWORK","_number":8,"created":"2022-05-09 17:25:23.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/81/835881/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/835881/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/835881/8"}}},"commit":{"parents":[{"commit":"db33bc0e72228711c70385be8287f5191a1bef5f","subject":"Merge \"Update Volume Delete rejection message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db33bc0e72228711c70385be8287f5191a1bef5f"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-03-30 16:36:21.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-09 17:24:30.000000000","tz":-240},"subject":"mypy: cinder/volume/volume_types.py","message":"mypy: cinder/volume/volume_types.py\n\nChange-Id: I89fc0f75d4839fedc4fc771e6d23b9526109cf4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/187142ccaef8e942ad9b34baa4e8ab9b8f42af9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/187142ccaef8e942ad9b34baa4e8ab9b8f42af9e"}]},"branch":"refs/heads/master"},"5dd65aff18c6f1b3c8fe13458f8ee54e9fefce2e":{"kind":"REWORK","_number":9,"created":"2022-05-09 22:01:04.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/81/835881/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/835881/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/835881/9"}}},"commit":{"parents":[{"commit":"db33bc0e72228711c70385be8287f5191a1bef5f","subject":"Merge \"Update Volume Delete rejection message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db33bc0e72228711c70385be8287f5191a1bef5f"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-03-30 16:36:21.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-09 21:54:38.000000000","tz":-240},"subject":"mypy: cinder/volume/volume_types.py","message":"mypy: cinder/volume/volume_types.py\n\nChange-Id: I89fc0f75d4839fedc4fc771e6d23b9526109cf4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5dd65aff18c6f1b3c8fe13458f8ee54e9fefce2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5dd65aff18c6f1b3c8fe13458f8ee54e9fefce2e"}]},"branch":"refs/heads/master"},"202035a54e689c790d3b75ade9cc634d542aa61a":{"kind":"REWORK","_number":10,"created":"2022-05-10 15:55:07.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/81/835881/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/835881/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/835881/10"}}},"commit":{"parents":[{"commit":"db33bc0e72228711c70385be8287f5191a1bef5f","subject":"Merge \"Update Volume Delete rejection message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db33bc0e72228711c70385be8287f5191a1bef5f"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-03-30 16:36:21.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 15:37:19.000000000","tz":-240},"subject":"mypy: cinder/volume/volume_types.py","message":"mypy: cinder/volume/volume_types.py\n\nChange-Id: I89fc0f75d4839fedc4fc771e6d23b9526109cf4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/202035a54e689c790d3b75ade9cc634d542aa61a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/202035a54e689c790d3b75ade9cc634d542aa61a"}]},"branch":"refs/heads/master"},"adfc4cc8acd0abd45f7257fd436bfcb6604dd526":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2022-05-12 13:48:08.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/81/835881/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/835881/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/835881/11"}}},"commit":{"parents":[{"commit":"db33bc0e72228711c70385be8287f5191a1bef5f","subject":"Merge \"Update Volume Delete rejection message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db33bc0e72228711c70385be8287f5191a1bef5f"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-03-30 16:36:21.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-05-12 13:48:08.000000000","tz":0},"subject":"mypy: cinder/volume/volume_types.py","message":"mypy: cinder/volume/volume_types.py\n\nDepends-on: https://review.opendev.org/c/openstack/cinder/+/841591\n\nChange-Id: I89fc0f75d4839fedc4fc771e6d23b9526109cf4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/adfc4cc8acd0abd45f7257fd436bfcb6604dd526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/adfc4cc8acd0abd45f7257fd436bfcb6604dd526"}]},"branch":"refs/heads/master","description":"Edit commit message"},"85e914656884971a18226bde89e28a672fdfe67f":{"kind":"REWORK","_number":12,"created":"2022-05-16 16:57:23.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/81/835881/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/835881/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/835881/12"}}},"commit":{"parents":[{"commit":"ef9a3d6299503607cc0deb0839261350a116f0f8","subject":"Merge \"Increase swap size to 4GB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ef9a3d6299503607cc0deb0839261350a116f0f8"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-03-30 16:36:21.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-16 16:56:51.000000000","tz":-240},"subject":"mypy: cinder/volume/volume_types.py","message":"mypy: cinder/volume/volume_types.py\n\nChange-Id: I89fc0f75d4839fedc4fc771e6d23b9526109cf4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85e914656884971a18226bde89e28a672fdfe67f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85e914656884971a18226bde89e28a672fdfe67f"}]},"branch":"refs/heads/master"},"f43bed1c6ff2fa9b3088fe815b98fa47501987ce":{"kind":"REWORK","_number":13,"created":"2022-05-19 18:34:02.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/81/835881/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/835881/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/835881/13"}}},"commit":{"parents":[{"commit":"be71720bb990b15d9e29813ee393077ff8e1b772","subject":"Merge \"Use modern type annotation format for collections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be71720bb990b15d9e29813ee393077ff8e1b772"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-03-30 16:36:21.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-19 18:33:29.000000000","tz":-240},"subject":"mypy: cinder/volume/volume_types.py","message":"mypy: cinder/volume/volume_types.py\n\nChange-Id: I89fc0f75d4839fedc4fc771e6d23b9526109cf4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f43bed1c6ff2fa9b3088fe815b98fa47501987ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f43bed1c6ff2fa9b3088fe815b98fa47501987ce"}]},"branch":"refs/heads/master"},"04b6700f48f0776f7bedbea7a26ff18070117c1e":{"kind":"REWORK","_number":14,"created":"2022-05-23 14:36:24.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/81/835881/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/835881/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/835881/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/835881/14"}}},"commit":{"parents":[{"commit":"be71720bb990b15d9e29813ee393077ff8e1b772","subject":"Merge \"Use modern type annotation format for collections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be71720bb990b15d9e29813ee393077ff8e1b772"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-03-30 16:36:21.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-23 14:36:17.000000000","tz":-240},"subject":"mypy: cinder/volume/volume_types.py","message":"mypy: cinder/volume/volume_types.py\n\nChange-Id: I89fc0f75d4839fedc4fc771e6d23b9526109cf4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04b6700f48f0776f7bedbea7a26ff18070117c1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04b6700f48f0776f7bedbea7a26ff18070117c1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
