)]}'
{"id":"openstack%2Fcinder~836252","triplet_id":"openstack%2Fcinder~master~I0dd59ed5d7e63ea7a5e80c6a0505e6e1f1f60045","project":"openstack/cinder","branch":"master","topic":"bug/1967686","attention_set":{"33473":{"account":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"last_update":"2024-10-12 13:35:01.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"35075":{"account":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"last_update":"2024-10-12 13:35:01.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2024-10-12 13:35:01.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2023-07-22 14:37:45.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"34807":{"account":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"last_update":"2023-07-22 14:37:45.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-14 07:59:44.000000000","reason":"Someone else replied on the change"},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2023-07-22 14:37:45.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-07-22 14:37:45.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2024-01-10 20:03:50.000000000","reason":"\u003cGERRIT_ACCOUNT_13425\u003e replied on the change","reason_account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}}},"hashtags":[],"change_id":"I0dd59ed5d7e63ea7a5e80c6a0505e6e1f1f60045","subject":"Deprecate [DEFAULT] use_forwarded_for","status":"NEW","created":"2022-04-03 15:37:22.000000000","updated":"2026-06-19 20:02:14.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"76ecb1cbc95d1873282d984690bae3af6729be0d","_number":836252,"virtual_id_number":836252,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-11 16:26:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":1,"date":"2026-05-26 11:12:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}],"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-03 15:37:34.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-03 16:11:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-03 16:49:30.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-04-03 16:54:11.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-03 16:55:42.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-04-03 17:58:16.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-04-03 18:08:21.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-03 18:09:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-03 18:29:03.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-04-03 18:54:36.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-04-03 19:03:38.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-03 20:23:02.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-03 21:27:04.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-04-03 21:54:01.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-04-03 22:46:30.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-04-03 22:59:35.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-04-03 23:13:43.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-04-03 23:59:30.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-04 03:35:35.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-05 17:40:38.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-06 00:34:27.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-04-11 14:49:58.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-11 23:16:39.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-04-11 23:18:36.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-05-02 03:29:08.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-19 10:21:01.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2022-07-15 15:12:19.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2022-07-16 16:47:57.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-07-16 17:30:54.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-07-18 08:28:49.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-09-16 09:07:34.000000000","updated_by":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"reviewer":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"state":"REVIEWER"},{"updated":"2023-01-20 16:12:56.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2023-05-19 14:43:16.000000000","updated_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"reviewer":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"state":"REVIEWER"},{"updated":"2023-07-21 15:55:29.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-07-22 17:13:14.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-07-23 17:20:23.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2023-07-24 04:00:11.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-12-13 08:15:26.000000000","updated_by":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"reviewer":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"state":"REVIEWER"},{"updated":"2024-03-15 14:57:22.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-10-12 15:23:44.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-10-12 18:05:55.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-07-20 17:41:42.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-11 14:22:31.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-05-11 14:24:45.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-11 14:32:34.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-05-26 11:12:09.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-06-19 20:02:14.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"2c5f6dca71b819ff5ff691baebc4ce137e0f6842","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-03 15:37:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"260e8aa78f22bbd32c9dc04be697148d44e4e4a4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-03 15:37:34.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15704/","accounts_in_message":[],"_revision_number":1},{"id":"260dafcdd0c9897f2ea641de8e6e595f13ee8bc5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-03 15:39:15.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15704/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"5bfa0f752c2bff48a29cdfcbba868ad5976fc4bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-03 15:49:28.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ef77b8769d8e731e4584d1d55651302269dd2a50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 16:11:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3f92c0fc3214532b0ef06517ad2edf8 : SUCCESS in 17m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92d583da35bd4172974ab8f8f0ef4ca6 : SUCCESS in 19m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"67fdb0ef5d3f659a68c97c58e73e1cc3f85edc25","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-03 16:49:30.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/836252/2/iscsi/ : FAILURE in 58m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/836252/2/smb/ : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":2},{"id":"6aa1d270bbe2f816ca01c9f6cc2b59b246f961b9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-03 16:54:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/836252 : FAILURE in 55m 55s","accounts_in_message":[],"_revision_number":2},{"id":"135acdb8c5fbc73363b2a4734032ceb565c7a91a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-03 16:55:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/52/836252/2/check/iscsi-driver-alletra-6k/abc10a0/abc10a0.html : FAILURE in 1h 00m 26s","accounts_in_message":[],"_revision_number":2},{"id":"7359d6a5fffd36b92e8c75179261fbc01fcf71c6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-03 17:18:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/836252 : SUCCESS in 1h 19m 24s","accounts_in_message":[],"_revision_number":2},{"id":"44688d3451919b9bf96a580cece14fdc5973aae0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-03 17:58:16.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/52/836252/2/check/dsvm-tempest-cisco-zonemanager-job/e350077/ : SUCCESS in 1h 10m 26s","accounts_in_message":[],"_revision_number":2},{"id":"e68411f2db5300afdd7479f5b5ea72de91109750","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-03 18:08:21.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/9f08e1a777f84325a913d3448103faec : POST_FAILURE in 1h 54m 14s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/2cbcb1df45fb4b7497debed9ec0f1de0 : POST_FAILURE in 2h 14m 59s","accounts_in_message":[],"_revision_number":2},{"id":"df8ca06c98b5e54bd80ef8609dfe5f7e16fd7801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 18:09:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/255be9a428fa465595bf3ec412e538fc : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e93c351c3c7b40519390df877498de6e : SUCCESS in 13m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f47c8f855aa4d1c9fc9c5b8dff318ea : SUCCESS in 11m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57593b9b150d4ccd8b01452c506b843c : RETRY_LIMIT in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c44d009248d46d1bd95f270451c33fb : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/d76db807be1d4507a6451f588a193d4a : SUCCESS in 52m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f4153b8d71e646f6a35ed9d6921bd20f : SUCCESS in 1h 15m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90b9ecc67d754e2e81aff18e84162c58 : SUCCESS in 46m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5373eaf06d084ff7b6432630cf146f5f : SUCCESS in 7m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a47e2dd8c69642779b50938ba21d572d : SUCCESS in 15m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6c69539ed57f4dd596619c212ea61252 : SUCCESS in 6m 07s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5fa5fa809b0346b7a37c5503a4966837 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/95c3390c11334fbeaf6911b8ca30c4c4 : SUCCESS in 7m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9ffe7cf7d66b426cb1ce1a879cb39e55 : SUCCESS in 6m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e464e63ebc8247a9aafcd1f4d9fdacdf : SUCCESS in 44m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c22c04725b4c4eab916ed54dd4fcff34 : SUCCESS in 7m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2b1e18ec84124b7d8178de096f5a4f53 : SUCCESS in 1h 10m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b68fb99a273f410aa93c4b03af06ebea : SUCCESS in 1h 19m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4949fce1eace49608fb3ce64642fdfbd : SUCCESS in 1h 33m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f1249ba6f1b447cc9630130c4556995d : SUCCESS in 1h 02m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f0b1db78f16449ec84c52310b7630b74 : SUCCESS in 1h 20m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/88aa030bfc8e4cefa7a7ee055c41376a : SUCCESS in 45m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7c4b3547ff4c4b389ccbe89f6146abec : SUCCESS in 1h 20m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/40a3f575efcc42158a3886e1c8e466fa : SUCCESS in 2h 17m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3dd45273b5d4364a94932e3cb452eff : SUCCESS in 36m 38s","accounts_in_message":[],"_revision_number":2},{"id":"94363dc47e373e762e0c3e430ef38caf49ebfcf2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-03 18:22:52.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/52/836252/2/check/iscsi-driver-nimble/82e6ab4/82e6ab4.html : FAILURE in 57m 36s","accounts_in_message":[],"_revision_number":2},{"id":"330cb31bae01e8426d1c02e33c4379de8ada43f9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-04-03 18:29:03.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/03_04_2022_18_12_34\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"24bfe81739afa72c745b1145a7a43213101c3f93","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-03 18:54:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/52/836252/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/41a4a24/ : TIMED_OUT in 3h 03m 03s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/52/836252/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b986c1e/ : TIMED_OUT in 3h 03m 01s","accounts_in_message":[],"_revision_number":2},{"id":"1b435151784ae9ec7e61e4ca94d374351cd048b5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-03 19:03:38.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-52-836252-1/ns5_nfs : SUCCESS in 3h 25m 45s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-52-836252-1/ns5_iscsi : SUCCESS in 3h 24m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"f506d1bfdf6e065c241dcc9ac3a03e34ca9c8f2a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-03 19:18:07.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-52-836252-2/ns5_nfs : SUCCESS in 3h 25m 53s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-52-836252-2/ns5_iscsi : SUCCESS in 3h 28m 7s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"15765f998b146fd6ec48066be6840a5892cdd97b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-03 19:23:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/52/836252/2/check/iscsi-driver-nimble/52130e1/52130e1.html : FAILURE in 59m 22s","accounts_in_message":[],"_revision_number":2},{"id":"d36da1045022f11c582f9308814f8d3f0db3de32","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-04-03 20:23:02.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/52/836252/2/check/DellEMC_PowerFlex/447e3e7/DellEMC_PowerFlex/524 : SUCCESS in 1h 24m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"6f3ebadf1d5874ea7d071351ce42a9ee8382b15b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-04-03 21:27:04.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/52/836252/2/check/DellEMC_SC_FC/7c2f719/DellEMC_SC_FC/392 : SUCCESS in 1h 05m 05s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/52/836252/2/check/DellEMC_SC_iSCSI/cfc36d8/DellEMC_SC_iSCSI/413 : SUCCESS in 1h 51m 45s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"3d272c07b92244ddafa1dd6e35d1cb27520d6389","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-04-03 21:54:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/836252-2/HPMSA-iSCSI/ : SUCCESS in 51:41.71\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/836252-2/HPMSA-FC/ : SUCCESS in 44:52.69","accounts_in_message":[],"_revision_number":2},{"id":"c696640506f8b689329e4800e365b950dc471685","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-04-03 22:46:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/836252-2/Lenovo-iSCSI/ : SUCCESS in 1:04:01\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/836252-2/Lenovo-FC/ : SUCCESS in 57:41.46","accounts_in_message":[],"_revision_number":2},{"id":"e849c4bf53b75a208b77cac4f942947dca6fcf1a","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-04-03 22:59:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/836252-2/PVME-iSCSI/ : SUCCESS in 1:04:34\n- PVME-FC https://os-logs.seagate.com/dell-ci/836252-2/PVME-FC/ : SUCCESS in 1:02:27","accounts_in_message":[],"_revision_number":2},{"id":"ff5d7008d921dacaca3abb4f50af2b7bc05a12d7","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-04-03 23:13:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/836252-2/Seagate-iSCSI/ : FAILURE in 1:17:31\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/836252-2/Seagate-FC/ : SUCCESS in 1:05:28","accounts_in_message":[],"_revision_number":2},{"id":"1747be534c1fb788bab42d24681a8ce5e9cf242a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-04-03 23:59:30.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/52/836252/2/check/DellEMC_PowerMAX_FC/34d934e/DellEMC_PowerMAX_FC/409 : SUCCESS in 3h 09m 01s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/52/836252/2/check/DellEMC_PowerMAX_iSCSI/a054493/DellEMC_PowerMAX_iSCSI/411 : FAILURE in 4m 29s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"08c9df40ad2661aa2fced5ccd6d521dbc27b068f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-04-04 03:35:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/2/20220404091411/NEC-FC : SUCCESS in 3h 21m 05s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/2/20220404091411/NEC-ISCSI : SUCCESS in 3h 21m 20s","accounts_in_message":[],"_revision_number":2},{"id":"ba1f92765d30c71ac275b2f82275c85c8c3d1a9f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-04-05 17:40:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/52/836252/2/ : SUCCESS in 1h 58m 46s","accounts_in_message":[],"_revision_number":2},{"id":"54bbe0f6e0f829b5aa3ade7e199e8af1eaff166d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-04-06 00:34:27.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/2/20220406072608/NEC-V-FC : SUCCESS in 1h 53m 41s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/2/20220406072608/NEC-V-ISCSI : SUCCESS in 2h 08m 16s","accounts_in_message":[],"_revision_number":2},{"id":"9b8554b78d53ebd75a57d81c42b011a07cca0ff2","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-04-11 14:49:58.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC https://elab-os-logsrv.delllabs.net/52/836252/2/check/DellEMC_XtremIO_FC/138636d/DellEMC_XtremIO_FC/902 : SUCCESS in 1h 11m 27s\n- DellEMC_XtremIO_iSCSI https://elab-os-logsrv.delllabs.net/52/836252/2/check/DellEMC_XtremIO_iSCSI/c8f976e/DellEMC_XtremIO_iSCSI/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"10271df74787822936db5f65da00d29b0561ec42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-11 23:16:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a09f814f17b06b0322c60f75087403dca8468e14","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-11 23:16:39.000000000","message":"Patch Set 3:\n\nBuild Started","accounts_in_message":[],"_revision_number":3},{"id":"0d67a37af831faf9b9301e17d8fb539c67232239","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-11 23:16:41.000000000","message":"Patch Set 3:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15788/","accounts_in_message":[],"_revision_number":3},{"id":"722bd9ea0bf05db8923a932dd042ac1bac9f7bb4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-11 23:18:35.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15788/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"888d777537900d5c0b345805d74705b19d11f999","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-11 23:18:36.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/52/836252/3/check/DellEMC_Unity_FC/a919fa8/DellEMC_Unity_FC/593 : FAILURE in 1m 45s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/52/836252/3/check/DellEMC_Unity_iSCSI/d68c1c8/DellEMC_Unity_iSCSI/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"d00b6cb7d9aef26b7c98d773e8d69e6f77f39799","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-11 23:56:40.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1649718998416 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"61ceb12ccd507e6ca03563bacfb24d9cbbe91c18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 00:00:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1aad72c687a49b18df28394d6e1b3b3 : SUCCESS in 25m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ae1e50690ef42dcbf8a43a97abaf698 : SUCCESS in 23m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f2dbd5f184a8cad4c386ee5de254b01affd3ba47","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-12 00:13:08.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/836252 : FAILURE in 55m 04s","accounts_in_message":[],"_revision_number":3},{"id":"661fa921fa54c20b1faddedb2a1dad3502d88615","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-12 00:22:37.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/52/836252/3/check/iscsi-driver-alletra-6k/03ebbed/03ebbed.html : FAILURE in 1h 04m 39s","accounts_in_message":[],"_revision_number":3},{"id":"cb79c33a74369673bd6a570efcd1e1f95ac2a4a1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-12 00:50:51.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/836252/3/iscsi/ : SUCCESS in 1h 28m 35s\n- cinder-smb http://cloudbase-ci.com/cinder/836252/3/smb/ : SUCCESS in 1h 31m 40s","accounts_in_message":[],"_revision_number":3},{"id":"532201840dfb8b0cf5e3472c02057bfb371c1439","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-12 01:33:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/836252 : SUCCESS in 1h 22m 13s","accounts_in_message":[],"_revision_number":3},{"id":"2c222b51141b313eeec64b6a4948e4ae4845658a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 02:04:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b985fc1ac20247f1b91ead2feeeff803 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8bf0deb61374bd991eeace964be7678 : SUCCESS in 11m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/deda3cf7be8c49e88d962d690c82ab0a : SUCCESS in 11m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c20e0e706384b05a9ff1de478407183 : RETRY_LIMIT in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48b98e9dd5ce474597e9777328eeae17 : SUCCESS in 11m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/2a64842c23744695921c8908a54e112a : SUCCESS in 1h 04m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fb4d3f5044d34f5da9a07396599feb94 : SUCCESS in 1h 18m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e274e7173fc441195f0fcf8698737fc : SUCCESS in 39m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fceb6f839a54ac780f63b9e7f413c45 : SUCCESS in 8m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/51983ddeced14542a875b8eb46c28bc0 : SUCCESS in 16m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f070ac9d2f60427bb12ca7fa38e4c827 : SUCCESS in 6m 39s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7b06fd57fc8047d89be1c46f2e3b20d2 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/54eab5c5dcea4989917042819b7cb85d : SUCCESS in 6m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/146e0906a45249f8a1bb95d425345dd1 : SUCCESS in 7m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6d2a8f770ade4f6dbf7eb620e4571534 : SUCCESS in 37m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a6b610cc42924232bedb9af1cc951072 : FAILURE in 5m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cb1ced87dbac4d6695f4e57cb2a7f7ac : SUCCESS in 1h 19m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/133c8743543041c3951726e1e7473d50 : SUCCESS in 1h 24m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/68bb21afd4134978bcd21fbdf301cf23 : SUCCESS in 1h 36m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3b4b5f9ad49e4fe6a91242b28f3aee3d : SUCCESS in 1h 16m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/637c654a48144af2964122e192b5f486 : SUCCESS in 1h 16m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9b5e9bacae3a4aa68bf10c0656bc3e61 : SUCCESS in 50m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4004c5f9bd4044d9abac295eb108b5bf : SUCCESS in 1h 15m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6dc22f1333054070987f64cd24756db7 : SUCCESS in 2h 39m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/081d245beccf46369b456b7e1646d71d : SUCCESS in 52m 20s","accounts_in_message":[],"_revision_number":3},{"id":"637bdc36348c07bcbb0144c4be21bcb5d98b6b3e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-04-12 02:22:06.000000000","message":"Patch Set 3:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/12_04_2022_02_04_16\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"30e596ccf70829b9cbd2203a2272d0679359627e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-12 02:35:20.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/52/836252/3/check/DellEMC_Unity_FC/e2259af/DellEMC_Unity_FC/595 : FAILURE in 21m 22s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/52/836252/3/check/DellEMC_Unity_iSCSI/96ed4ab/DellEMC_Unity_iSCSI/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"f3e1465555546389b04155d0032b30f3b348ac6b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-04-12 04:04:32.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/52/836252/3/check/DellEMC_SC_FC/a7f63cf/DellEMC_SC_FC/447 : SUCCESS in 59m 33s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/52/836252/3/check/DellEMC_SC_iSCSI/eb91134/DellEMC_SC_iSCSI/468 : SUCCESS in 2h 00m 07s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"9b6a20f72a0e6a4257b2315da0b34cdb42a08440","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-12 04:31:17.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/9f9f71d346b049839a59de8bad02f9db : POST_FAILURE in 2h 39m 33s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/ae082e4771ab4518a1b160d95c77cad9 : TIMED_OUT in 3h 09m 16s","accounts_in_message":[],"_revision_number":3},{"id":"34912ddc123429d9125b53b304b39e807feafb28","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-12 05:56:55.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/52/836252/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/7ea1214/ : TIMED_OUT in 3h 02m 53s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/52/836252/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/591bc9f/ : SUCCESS in 1h 28m 00s","accounts_in_message":[],"_revision_number":3},{"id":"7e42f44e211be2e707a465afac883ada77d78023","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-12 06:45:23.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"d1963abace7ac999e9d023d9c22895aa866d032f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-12 06:45:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"be9e7fdc523f832ec872ef38ce555d8c62259ca9","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-04-12 12:52:08.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/52/836252/3/ : FAILURE in 1h 51m 29s","accounts_in_message":[],"_revision_number":3},{"id":"e462b38ca0cc76b39b9fcfd2b80f0b3d57d391b8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-04-12 14:35:40.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/52/836252/3/check/DellEMC_PowerFlex/b8e3660/DellEMC_PowerFlex/575 : SUCCESS in 1h 21m 21s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"40b09d695aa4d88d9bb501e0eb351f2ef792e4db","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-12 16:08:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/52/836252/3/check/dsvm-tempest-cisco-zonemanager-job/c8394b2/ : SUCCESS in 1h 11m 35s","accounts_in_message":[],"_revision_number":3},{"id":"d21749c53cdfc4bfe99d1604a59b5c7144c8c6f6","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-04-12 21:00:36.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"3a8790f514ac80e0ec04d7595a81c33a079817fb","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-04-12 21:29:06.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"31a4b7fb01c04b3fc9697b37e03e8cb1c65e37e1","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-04-12 21:58:48.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"4a129eb4ac5771ad636647e13c061934021f06f0","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-04-12 22:07:07.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"a6943c0d74fa76e1e8163311022e6ac9fcf53794","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-04-14 14:44:24.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/3/20220414211909/NEC-V-FC : SUCCESS in 1h 53m 28s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/3/20220414211909/NEC-V-ISCSI : SUCCESS in 2h 25m 12s","accounts_in_message":[],"_revision_number":3},{"id":"3a089a6f970ce0e71c6ed43db64b3d05b4a0f5b6","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-04-16 08:54:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/3/20220416143212/NEC-ISCSI : SUCCESS in 3h 17m 59s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/3/20220416143212/NEC-FC : SUCCESS in 3h 22m 29s","accounts_in_message":[],"_revision_number":3},{"id":"7708f038d5c12f56e54d370cf7fa5016c1099cc9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-02 03:29:08.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/54274a44ccbc4c5694605355ba04e7e2 : FAILURE in 18m 10s","accounts_in_message":[],"_revision_number":3},{"id":"eb0061fe655abc02014908118a4d7d146705b10c","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-05-19 10:21:01.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e57d5b391165c3ce30de0e574b3d8c687bbfaaaf","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-07-15 15:12:19.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a16c120e0da64a266d7248366dd988c5c1cf77ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-16 14:13:58.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"504c665967c6e5623386073385ba21112e56c3ff","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-16 14:14:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5665ed881fa56caca4b760283b84a7fd5c5c9ab2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-16 14:14:20.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"1904b782054ee40dd84a270fcfe18c5ca8dd5e77","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-16 14:31:17.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1657980846720 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"13e8860ed0417d23b388dbaed92dd0064f2883f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-16 14:39:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a4ea3616d3544369af102236be597a2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/654238c251884fbfbfa2bc3f2ab2a2a9 : SUCCESS in 23m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/93bd2bce138448cc963e87b56ab51f77 : SUCCESS in 23m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c92b7f709d21a9519911a49a8d589beb01704961","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-16 14:48:23.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1657980871797 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"ee8bb22ef064b959b0bb174f60d6798fc1c1372c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-07-16 14:57:07.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e3b9cec318b449b3ae09264ab965cb80 : SUCCESS in 41m 22s","accounts_in_message":[],"_revision_number":5},{"id":"01ebdaf3519499620ebd848a6f37151bbb4b9346","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-16 15:10:31.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/52/836252/5/check/dsvm-tempest-cisco-zonemanager-job/4a2ac97/ : FAILURE in 51m 45s","accounts_in_message":[],"_revision_number":5},{"id":"2b15e305b88ce98fb5cd80ecbe738fbe74bc54c9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-16 15:52:32.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/52/836252/5/check/tempest-dsvm-full-focal-py3/0bfc435/ : SUCCESS in 1h 19m 15s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/52/836252/5/check/devstack-plugin-ceph-dsvm-focal-py3/0992f07/ : SUCCESS in 1h 35m 54s","accounts_in_message":[],"_revision_number":5},{"id":"9b223fa93107c3633ab2de861d3610aba1d4e80e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-16 15:52:42.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/836252/5/iscsi/ : FAILURE in 1h 37m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/836252/5/smb/ : SUCCESS in 1h 12m 32s","accounts_in_message":[],"_revision_number":5},{"id":"41deca5cff9f888826c390ff56b067cae83dcfe7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-16 16:07:04.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/52/836252/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b2952a8/ : SUCCESS in 1h 23m 06s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/52/836252/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/25a88c0/ : SUCCESS in 1h 51m 33s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/52/836252/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/02cd27d/ : FAILURE in 33m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c19619fc84818e480cde2702bf36e3f68dfdc047","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-16 16:28:20.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/836252 : FAILURE in 2h 12m 21s","accounts_in_message":[],"_revision_number":5},{"id":"6cef3e000fdd6bf10ef7fcb75b457a9c5f9bf85f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-16 16:47:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/364a9d06fbe94aa5ae27f4049ec9b693\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43e7c96d0cd6472c9b19737409c80cd5 : SUCCESS in 8m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a64d0da2e9064beb82cee73b6afe98be : SUCCESS in 11m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/613f32bdba754656b2952c1d1fbad8ff : SUCCESS in 13m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30d84dbf90a743128e7efe0bb25f9bc3 : SUCCESS in 12m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de673bbc96ee4e56a328b8227f0c494e : SUCCESS in 14m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/fc30214ca5574955ba493ece8aec8983 : SUCCESS in 59m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4951e0e4f8314725a4df4dfea3bb875d : SUCCESS in 1h 28m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5147dc3a833945d4b8a67f4a359353b8 : SUCCESS in 1h 00m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5db7141026f4e13a0210c09c11a5af9 : SUCCESS in 8m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2150242a48404a9f805b67d13863a7d8 : SUCCESS in 15m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0c1725b9e65843ddbe8d8ab0c45c1a55 : SUCCESS in 6m 00s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/51a743f8ac7e4b23bba53e30fa3669ad : SUCCESS in 9m 08s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c56a65123d7547b292076cfb54aa4072 : SUCCESS in 7m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/58ea6dc7719d4b589099eaecaec0a7af : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3299b0cdd836446facefa2d2c43b13db : SUCCESS in 49m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5caea7a13e054d699363c993cee208ec : FAILURE in 6m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/44fa9659d52a4e4e88c0795f9d9bb98b : SUCCESS in 1h 31m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9f4d1c09f1304303b248fbdc05bb3511 : FAILURE in 1h 29m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5fe4ea14a752442c88d92906cc11451c : SUCCESS in 1h 59m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fc10261a759d4718b8302e2d23046dab : SUCCESS in 1h 10m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fbc1e1678ec54150b11e98f686c521c9 : SUCCESS in 1h 36m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f2348870b03f4effb3ca204d34b3b2f6 : SUCCESS in 56m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a4d30c4a05274e5b95af04b78dbe16d9 : SUCCESS in 1h 39m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b31a4cf6757448a5947ef82e6aacd463 : SUCCESS in 2h 24m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4f4cdb91b304059af6a8ed98521145d : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":5},{"id":"b22cbac2a26f7e6894d61577493d80974f80c001","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-16 16:47:57.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/52/836252/5/check/DellEMC_PowerStore_NFS/ea53474/DellEMC_PowerStore_NFS/137 : FAILURE in 1s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"6a5b3a4b5a8bc0e5c83e4f2a3c41f4cca9f7ea7e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-16 17:30:54.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/52/836252/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/24e7b85/ : SUCCESS in 42m 05s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/52/836252/5/upstream-check/cinder-tempest-plugin-ontap-nfs/d08df4b/ : SUCCESS in 41m 09s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"a93ec2d3da8d54bb635c6c3ab6f05d4f6e176ce7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-16 17:42:30.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-52-836252-4/ns5_nfs : SUCCESS in 3h 26m 16s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-52-836252-4/ns5_iscsi : SUCCESS in 3h 28m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"cc26a923d303196cb51e5265b9b5e2ec5e78215b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-16 18:05:20.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-52-836252-5/ns5_nfs : SUCCESS in 3h 26m 28s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-52-836252-5/ns5_iscsi : SUCCESS in 3h 28m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"435050a851bb73778c52cc4d5e89a7a1ae4a423e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-16 18:21:09.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/52/836252/5/check/DellEMC_PowerFlex/8a777ab/DellEMC_PowerFlex/1005 : SUCCESS in 1h 33m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"69d4ad09a1959885c9b4d9aa5e0db5aa014f18a5","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-07-16 19:16:00.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"29870176054f176faf62e35d0a896844fa84f9e8","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-07-16 20:27:10.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/52/836252/5/check/DellEMC_SC_FC/12a6508/DellEMC_SC_FC/789 : SUCCESS in 49m 15s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/52/836252/5/check/DellEMC_SC_iSCSI/8331bd0/DellEMC_SC_iSCSI/811 : SUCCESS in 1h 45m 22s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"87393becb9e4f01d077369f777dea63eacd20f9c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-16 22:14:11.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/52/836252/5/check/DellEMC_Unity_FC/6918665/DellEMC_Unity_FC/1166 : SUCCESS in 1h 11m 26s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/52/836252/5/check/DellEMC_Unity_iSCSI/ffd40e4/DellEMC_Unity_iSCSI/1090 : SUCCESS in 1h 52m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"bfc9daeb4694ba22aefc82c06e36d1352b2c8b9c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-18 01:20:04.000000000","message":"Patch Set 4:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":4},{"id":"880389c3e56fa8c356b9b3356bf4ce5756bb8704","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-18 08:28:49.000000000","message":"Patch Set 4:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/125","accounts_in_message":[],"_revision_number":4},{"id":"f8214ae5ebe61bae872823fee87ffca4340728b6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-18 08:31:29.000000000","message":"Patch Set 4:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/128","accounts_in_message":[],"_revision_number":4},{"id":"50e599b4ca8df51a62177e64b46b60e664b94211","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-18 08:40:20.000000000","message":"Patch Set 4:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/129","accounts_in_message":[],"_revision_number":4},{"id":"35b4ca7d849dac274a41bb8612be50c0cee1e99c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-23 05:20:16.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/5/20220723120012/NEC-V-FC : SUCCESS in 1h 49m 24s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/5/20220723120012/NEC-V-ISCSI : SUCCESS in 2h 20m 00s","accounts_in_message":[],"_revision_number":5},{"id":"9ec98313392d4670f4f276d17997dfce8bf0bec3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-07-28 01:23:48.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/5/20220728064954/NEC-FC : SUCCESS in 3h 29m 35s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/5/20220728064954/NEC-ISCSI : SUCCESS in 3h 33m 51s","accounts_in_message":[],"_revision_number":5},{"id":"185fcb3404020a144aa748c66d07927cd8df6aac","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-08-22 13:28:08.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c8a19df99efb93f500242685d8567c091f4d3748","author":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"date":"2022-09-16 09:07:34.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ee21f8eb008b8e2ff6aba0a5d34d79079c617ed1","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-01-20 16:12:56.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"07222da2932f63bbbb0835ddffa181e38187c6be","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-05-19 14:43:16.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8d1af5883bc4845431f9ae3dff016e4bdc5f6d71","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-07-21 15:55:29.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cb9e33d857cbc779f7e635eff5a4c986413f9c43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-22 14:37:45.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"f599535cfa077b90798b7483ae9e51769a0adf46","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-22 14:38:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3dfaeed4ce8ddd2e53ef073be581734ed8c11001","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-07-22 15:04:11.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/52/836252/6/check/iscsi-driver-alletra-6k/653055d/653055d.html : FAILURE in 24m 30s","accounts_in_message":[],"_revision_number":6},{"id":"e905ff5bd01f4b3354786cc91fcfdbc05b447429","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-22 15:05:05.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e6a7a50aa094b0aa01a06ffd957b463\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/68a9eba8ab33454fb8c03a1d2f7a285e : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b72f86ff1c054ba488cdbfc3723f327a : SUCCESS in 25m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/efc1b240451a470c8cf3dd690f1dfe87 : SUCCESS in 14m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4aeea60ce01412b39fd9bcc9dc0990cc152e9bad","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-22 15:13:51.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/836252 : FAILURE in 34m 29s","accounts_in_message":[],"_revision_number":6},{"id":"2e286d7ac19571b63b116a9cf81d8dca1e9b861d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-07-22 15:17:59.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/52/836252/6/ : FAILURE in 32m 09s","accounts_in_message":[],"_revision_number":6},{"id":"87352c6443c843c009f931029f29697ee5cb2c91","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-22 15:24:48.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-52-836252-6/ns5_nfs : FAILURE in 46m 33s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-52-836252-6/ns5_iscsi : FAILURE in 46m 22s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"056effe959f3cb9f2fa85a5a55ce51beae6c0223","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-22 15:54:46.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/36999abe2ec64246a770e54768eb59e2\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/5187dd7e13a7413a96d346164ea61b2e : FAILURE in 1h 14m 24s","accounts_in_message":[],"_revision_number":6},{"id":"4e78f48f560ba9ec87773c3264438332a23ab203","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-22 16:04:43.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/45744c70cc0a41c5a40fdfdb6f5cb93d : SUCCESS in 1h 25m 01s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/14d5e5b685644651ac52d628f25b9ac7 : RETRY_LIMIT in 55m 53s","accounts_in_message":[],"_revision_number":6},{"id":"b032ffbd970a5b05325dea1349c1e399796b5eb8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-22 16:15:52.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/52/836252/6/check/DellEMC_PowerFlex/f9c5551/DellEMC_PowerFlex/303 : FAILURE in 1h 37m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"1374ea01716be7a4b0cf4dded6ad9c9b142c52ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-22 16:41:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae07000ea8584f249b5b624c082c8480\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb2fc02067b4d6580c2be12050d8731 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9956f7f1b0f4fcb87193c6f5176c24c : SUCCESS in 11m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c267b6961e5b4c509626034801a5b710 : SUCCESS in 10m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/163269d4e1fa41afa6db45bfdfce9950 : SUCCESS in 11m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6796a95ea5864be6936eeee47bb1ff9d : SUCCESS in 14m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/66e9c93468674176b6e8f4dc148327a3 : SUCCESS in 43m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a73caf1637c94391b6523fa883853acd : SUCCESS in 1h 32m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd7f7009c4394719bbc13ad7284cbea2 : SUCCESS in 51m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43a2c958a49f4a0d9265d0435a7ddcd2 : SUCCESS in 6m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7c40f247dba74e2abb1b0f1da5946e74 : SUCCESS in 15m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5ad39ec589f747419fbaef9f7e89466c : SUCCESS in 5m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4da82d7a35ba4bd4ba9001662573d3f4 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/46c30e6ebad2415aaaa6b666390c86b7 : SUCCESS in 6m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/93ceab82b3e74cce9560741d97f7b8bf : SUCCESS in 7m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/acc959f671394c879d90902f003c677c : SUCCESS in 34m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6b24e90624c845c6a5cd3ba866d68546 : SUCCESS in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4cdb44927c1b4ec48948421bc76834d7 : SUCCESS in 1h 49m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e57afd4753e94be18ba8d6abb9d17297 : SUCCESS in 1h 42m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a923a767f8ec40d4bf95e195817afb7b : FAILURE in 2h 02m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a25b5203b22d4809b8e1f9160d334441 : POST_FAILURE in 9m 22s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/035c410c4bd64bf39a76344ef3162301 : SUCCESS in 1h 02m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ab979d94ccca4d999b1dcd5496de791c : SUCCESS in 1h 34m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ee91e6ad76294807874013ae57a39a36 : SUCCESS in 42m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a912631ded04478a866760a891facbc7 : SUCCESS in 1h 41m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c7a375287d0a4950a9305610a3a42a0a : FAILURE in 9m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/24060cea20474920923ece3c773e61a0 : SUCCESS in 1h 26m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7293f4fd3a44072aa3310729f3fd070 : SUCCESS in 1h 00m 10s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/d3ac3c16d6f04b1f8d07d142211d8e69 : TIMED_OUT in 31m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4d8beba3baf4dc56f6898c9c569e9413685af072","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-22 16:45:14.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/52/836252/6/check/DellEMC_Unity_FC/c57ce5f/DellEMC_Unity_FC/457 : SUCCESS in 1h 50m 44s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/52/836252/6/check/DellEMC_Unity_iSCSI/bdbee5c/DellEMC_Unity_iSCSI/457 : SUCCESS in 2h 07m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"8f66969669bcf86741a3bf7663e35ad83141ce13","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-22 16:56:22.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/52/836252/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/9685c25/ : SUCCESS in 1h 53m 42s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/52/836252/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/1fb73a7/ : FAILURE in 2h 16m 57s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/52/836252/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/dc6dc59/ : SUCCESS in 2h 06m 03s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/52/836252/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/2f0cdbf/ : SUCCESS in 1h 58m 02s","accounts_in_message":[],"_revision_number":6},{"id":"9f06bf574de10fc5d692d874f2d2e1f56f2083d9","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-22 17:13:14.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-52-836252-6/FC/ : SUCCESS in 2h 35m 10s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-52-836252-6/iSCSI/ : SUCCESS in 2h 9m 53s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"575ac0a21ff3fc885a5b8d590181c495a8e5a6ac","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-22 17:35:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1636/","accounts_in_message":[],"_revision_number":6},{"id":"7008a7a11dbdc88be6024b7f3d2090ba9298d247","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-22 17:42:56.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/6/20230722233748/NEC-V-ISCSI : SUCCESS in 3h 03m 52s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/6/20230722233748/NEC-V-FC : SUCCESS in 3h 05m 05s","accounts_in_message":[],"_revision_number":6},{"id":"4f1faff2c6e7d8614d875a894fba05fbd01680be","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-23 11:26:54.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/6/20230723121855/NEC-ISCSI : SUCCESS in 4h 07m 21s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/6/20230723121855/NEC-FC : SUCCESS in 8h 07m 56s","accounts_in_message":[],"_revision_number":6},{"id":"4ee36edc26738abfcf3555ae974b3586e9b416e9","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-07-23 17:20:23.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/52/836252/6/check/tyds_test/a760d34 : FAILURE in 10h 00m 03s","accounts_in_message":[],"_revision_number":6},{"id":"7b29267dfe708dbfa5fd73568b511e7ecb958e31","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-24 04:00:11.000000000","message":"Patch Set 6:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/52/836252/6/fc : SUCCESS in 0h 45m 08s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/52/836252/6/iscsi : SUCCESS in 0h 56m 22s","accounts_in_message":[],"_revision_number":6},{"id":"9cf2c1b697eb1c6d448bfca261fd78e7df602f61","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-25 23:35:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/406/","accounts_in_message":[],"_revision_number":6},{"id":"7e8611f6999cb978f477e5efbc0b9acc1f7d2dc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-11 12:58:12.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"4a59595ae605481e0e495b9080c81a8c7280ad6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-11 12:58:19.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"f433d92c1e48e55a7b289fb8f1ea055c14a74acb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-12-11 12:58:26.000000000","message":"Patch Set 7:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16260/","accounts_in_message":[],"_revision_number":7},{"id":"bb622346e71143b35641b92056710d65acd1160a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-12-11 12:58:29.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16260/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"6e5b363811a579ad8e2006971e6ecd298aab7fd3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-12-11 13:21:02.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-52-836252-7/ns5_nfs : FAILURE in 1m 12s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-52-836252-7/ns5_iscsi : FAILURE in 1m 12s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"3c1bcbd5ac9ac9890e044e3504ba0ac43e0386da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 13:21:26.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5753b4586d02420094601ba4cd7fef2f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca62206c82614d63a4f9d714ce1d7766 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5bdd914b2dbe493e9e87ae61fee6ab16 : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0c980955037e492182cf41bfea90d862 : SUCCESS in 21m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d4d0885031557fd599a5174cf6e600d15f093198","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-12-11 13:43:37.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-52-836252-8/ns5_nfs : FAILURE in 1m 10s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-52-836252-8/ns5_iscsi : FAILURE in 1m 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"5efc4b3c7552d128e9d644e35132286839880efe","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-12-11 13:59:26.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/836252 : SUCCESS in 53m 27s","accounts_in_message":[],"_revision_number":8},{"id":"0baf5bd651020f44100bb6fc2d187eaa343de6e9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-11 14:22:03.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/52/836252/8/check/DellEMC_Unity_FC/0412cbb/DellEMC_Unity_FC/1081 : FAILURE in 1h 09m 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/52/836252/8/check/DellEMC_Unity_iSCSI/cac761f/DellEMC_Unity_iSCSI/1096 : FAILURE in 33m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"49bc1f044cce69791412067548268158e04b8900","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-12-11 14:52:03.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/fbf763be997c4c80a0087a6c3c48d926\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/65cb9d33f498460397099023b3ece0a8 : FAILURE in 1h 24m 43s","accounts_in_message":[],"_revision_number":8},{"id":"c2a951420a64dcf8ec257e539cd68b12d2b928b9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-11 15:01:41.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/52/836252/8/check/DellEMC_PowerFlex/5ff45b8/DellEMC_PowerFlex/1036 : SUCCESS in 1h 46m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"94f93bf87ff95cd75cf3b80657a1758f883151ad","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-12-11 15:11:38.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/52/836252/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/dd79562/ : SUCCESS in 1h 53m 08s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/52/836252/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/592d685/ : SUCCESS in 1h 49m 09s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/52/836252/8/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/179a2a5/ : FAILURE in 2h 11m 51s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/52/836252/8/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/0f535d6/ : SUCCESS in 1h 47m 47s","accounts_in_message":[],"_revision_number":8},{"id":"d238299706b0c6142869964084d2efac09db7808","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 15:18:03.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d262eb5302d042978616f059f08d9dd2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72fb97103c3d468fb207be59ebbfdbf3 : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef47fe907c4644be9c823d9941f1f353 : SUCCESS in 11m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/016b06c1d9704c859852e4baf21e87d6 : SUCCESS in 11m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67384742564f4515b375fea05d379495 : SUCCESS in 13m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/ed30ea3292c049deb4cde12aeb82ee4f : SUCCESS in 1h 00m 51s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e8ecfcff7b814ec3a67d5ed7008b5fa0 : SUCCESS in 58m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b90b6b66f8d84250b8d232f0c3aac11f : SUCCESS in 1h 29m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a4f6d315113433ca444c6e9051de6e7 : SUCCESS in 1h 02m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2187c25b716640f6a30e2bf2d18d49e3 : SUCCESS in 8m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/150f4747e56c4ca48793e0c651eabc06 : SUCCESS in 17m 16s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/50907d7f4c99485985d8286313fc20df : SUCCESS in 12m 43s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/af57185c048b4fbeb284a80c6c55c05a : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7872d89d806140c082f93d9b1d955aa2 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eaf356aa62e545cfb231432737d69bff : SUCCESS in 6m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/578b777ad04f4cf0ad37951a643b4023 : SUCCESS in 5m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0af77793fb304261ab08f11b0cdc8ccf : SUCCESS in 39m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/10abcfa493354714a3347e5add54135d : SUCCESS in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2236e53ba6b84b72a48ed06a4efa14ea : SUCCESS in 1h 54m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1fb96a0c9b2b40e297d731544c78bc8e : FAILURE in 2h 06m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/00dc3213ef7a44aca1871798a239b5ef : POST_FAILURE in 2h 11m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0d0ac6c57bec4331886c6297e1fb31d9 : FAILURE in 1h 57m 00s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/76ef0e80a8424b76b0808a513e435f95 : SUCCESS in 1h 08m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9a1c92b8da9144db8fecfc86528c4fb0 : SUCCESS in 1h 38m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f5333380e39b4e85b3bc97545891c3d4 : SUCCESS in 44m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/799de40171284d38baf719163f478cb6 : SUCCESS in 1h 10m 26s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/553f3e47f29342c0ad99ad0557ccb8e0 : SUCCESS in 1h 38m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e18cd183433c467a8af72bc8799d0401 : SUCCESS in 1h 35m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b910953b19d14b93a9fa6e59b3e9bfcb : SUCCESS in 1h 09m 41s","accounts_in_message":[],"_revision_number":8},{"id":"d6e46470fa1ae41fd2ebbae7fe26e04a13234fb7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-12-11 16:16:11.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/250/.","accounts_in_message":[],"_revision_number":8},{"id":"60e5b8fc68b4fbf7170b27dd9eb397a4f1810b27","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-12-11 19:33:38.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/8/20231212011650/NEC-V-ISCSI : SUCCESS in 2h 56m 56s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/8/20231212011650/NEC-V-FC : SUCCESS in 3h 16m 45s","accounts_in_message":[],"_revision_number":8},{"id":"91773d85aec76c814847c299f92907ae5e13dbce","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-12-11 20:22:19.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-52-836252-8/FC/ : UNSTABLE in 2m 35s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-52-836252-8/iSCSI/ : FAILURE in 6h 25m 11s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"9e03016aaa496d12fe29b48e247409cc24235a55","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-11 21:00:56.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d8920ad99ec14858a6c4446b4182a6d2 : SUCCESS in 1h 07m 14s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f3d8437b002945f49e91982efd9db380 : FAILURE in 1h 10m 27s","accounts_in_message":[],"_revision_number":8},{"id":"31a73c061357e95293b227eaad491ce6c3cca846","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-12 15:13:38.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"49acb0b54467a3ae17688cd8afed8bf856438ad2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 15:39:19.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8534dcfd1dd440cb880a620c38a81f5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/afe5e01754134863beb4a2cb51b25ecc : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/63c3bda22bc64bf18e76835dc97475e7 : SUCCESS in 23m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6048e4d52ae14ffeb6f7aecb998087af : SUCCESS in 13m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b6ccc85d5c822b58ef4c799fd38c45a2549e77f2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-12-12 16:39:35.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/7461344e4b414a388e71b1ff10b4e847\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/9375508751fa42cf971384c37acc1ec9 : FAILURE in 1h 23m 02s","accounts_in_message":[],"_revision_number":8},{"id":"c911a7745b023d1375b6e8e8336834267a036d94","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-12 16:45:13.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7670815808b142418a10c76b37b243d9 : SUCCESS in 1h 11m 04s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/70a7beede6bc4c278c8df69e3552bfd5 : SUCCESS in 1h 05m 34s","accounts_in_message":[],"_revision_number":8},{"id":"a4120fe9cdb4011058887549fff11823db2d870c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-12-12 16:54:24.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/270/.","accounts_in_message":[],"_revision_number":8},{"id":"63d2260b529bfae4ca8f45d49fbdd0d21df942f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 17:47:11.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4841e9d5c9e4c13b278598395a994c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f046f642c35b40df90019eb8bc1854b2 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e033b18a17a466cb0a7c5f33fb48027 : SUCCESS in 11m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/48fbde1bf9e04320be0c339769a98984 : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48ae0fb008144d0cb3b31147929a55cb : SUCCESS in 14m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/04a72c530ac94ba9b6d17dc687dee57d : SUCCESS in 1h 07m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b6e53719aa414051b9e878081e882ea8 : SUCCESS in 59m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5813a47c18b848d590e6ad929cedc6ae : SUCCESS in 1h 24m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d87389281a7f464fa4210aad42384430 : SUCCESS in 1h 05m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fae03752366414ab11bc8c432b7a627 : SUCCESS in 9m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/26061df20a91422c9944b80babec1c2c : SUCCESS in 14m 59s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5d0cdc7ad0a04d0bb65d464d8ed0e705 : SUCCESS in 10m 46s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d893a5bd12cf49f8a021efbf79103314 : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/109fb569c1a24c6db3abceb2295a5c02 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e8c05df652df41cba4ffb6bfbbb524fa : SUCCESS in 6m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5e6efeba66eb43719f0c54bbb83f759a : SUCCESS in 5m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b0afc8d077994e4f9ee20877fb6ea827 : SUCCESS in 55m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a7dce939b18f49349e426005aa5fb6f0 : SUCCESS in 7m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fe570953a0214d7e80827ce3cc954cff : SUCCESS in 2h 05m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/06d9a145e35e42358aac7f2dc24c6abb : SUCCESS in 2h 05m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/af4f6c6495464555bfd8a70abff7923f : SUCCESS in 1h 52m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/233ba0b8527941acb06e9651e19d2102 : SUCCESS in 2h 09m 51s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2e1b6306e59249b3be2e6c7df7aa61ae : SUCCESS in 1h 06m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/58c7ffa88f5b481786c58f9d8281af39 : SUCCESS in 1h 21m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/71694e00e5b742dbaf8874c4ab54375b : SUCCESS in 43m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c9110cf0f9994dfb8e05270d87882baf : SUCCESS in 1h 31m 55s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/87a88ca72e9d4bbeac3a8aa1045d762a : SUCCESS in 1h 17m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7c9faf4111014063b4bb4085f20d8971 : SUCCESS in 1h 25m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5923e258c8f1435ca4f7afa38788ee2b : SUCCESS in 52m 03s","accounts_in_message":[],"_revision_number":8},{"id":"7259bb0b08cbc92649b2f77db9b8588804682baf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-12 18:53:04.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/52/836252/8/check/DellEMC_Unity_FC/827cba4/DellEMC_Unity_FC/1088 : FAILURE in 1h 05m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/52/836252/8/check/DellEMC_Unity_iSCSI/9e76264/DellEMC_Unity_iSCSI/1111 : FAILURE in 37m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"0a40c769b955f1e246fe5d836126f9859ed560d7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-12 18:59:43.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/52/836252/8/check/DellEMC_PowerFlex/5771d00/DellEMC_PowerFlex/1042 : FAILURE in 1h 12m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"65270cf7dae5e27967af4d04d95055bbda5a7d72","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-12-12 19:21:04.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/52/836252/8/check/DellEMC_PowerStore_iSCSI/88ef614/DellEMC_PowerStore_iSCSI/463 : SUCCESS in 1h 00m 30s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/52/836252/8/check/DellEMC_PowerStore_FC/bd03840/DellEMC_PowerStore_FC/466 : SUCCESS in 1h 24m 59s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/52/836252/8/check/DellEMC_PowerStore_NFS/d80a6f6/DellEMC_PowerStore_NFS/472 : SUCCESS in 1h 02m 04s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/52/836252/8/check/DellEMC_PowerStore_NVMeTCP/c202ca7/DellEMC_PowerStore_NVMeTCP/456 : SUCCESS in 1h 33m 35s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"b30bce15d89768028dc88b080acfdba1c5dbdd9e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-12-12 20:04:13.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/52/836252/8/upstream-check/cinder-tempest-plugin-ontap-fcp/13a2b5f/ : SUCCESS in 1h 10m 05s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/52/836252/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/809dad8/ : SUCCESS in 1h 48m 08s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/52/836252/8/upstream-check/cinder-tempest-plugin-ontap-nfs/3c85feb/ : SUCCESS in 1h 45m 51s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/52/836252/8/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/5dc4234/ : SUCCESS in 1h 51m 48s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"920ae4ca8f0f88b19a8bbfad171239407c15799b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-12-13 07:14:59.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/8/20231213115054/NEC-ISCSI : SUCCESS in 4h 03m 02s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/52/836252/8/20231213115054/NEC-FC : SUCCESS in 4h 24m 02s","accounts_in_message":[],"_revision_number":8},{"id":"08c1c775e746de2dfd090dd425ec89b5c2eed20b","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-12-13 08:15:26.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a02fd13d5ed63d279650eaf775c9c9698d6fad4a","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-12-13 19:00:16.000000000","message":"Patch Set 8:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/52/836252/8/fc : SUCCESS in 1h 00m 42s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/52/836252/8/iscsi : SUCCESS in 1h 11m 07s","accounts_in_message":[],"_revision_number":8},{"id":"bdb986d9a874ace993f386c0fd8fd31156311f20","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-12-24 15:25:36.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/52/836252/8/ : SUCCESS in 1h 30m 12s","accounts_in_message":[],"_revision_number":8},{"id":"ee7ae55816f6fe5325ba6fe1d78c3ed76b1cd5f9","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2024-01-10 20:03:50.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0b0b528f5516da243f7d37079a84b3bf367920d3","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2024-01-19 15:34:31.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"811362e21521ce89181d124edec13e5e934ecd49","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-03-15 14:57:22.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"8d00f69904592633414d3837c399807cdad5ea38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-12 13:28:09.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"380c69d5d22694ac810b2800024d9b21c6fc126e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-12 13:35:01.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":10},{"id":"2d2871859b090f82d3b88ce214352c60aad4c74b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-12 14:04:34.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fe7816a6c47499faa20c4b9aec11ca1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6dcc2de0c7d468a8c1b5f642280c38b : FAILURE in 21m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/74eca5b643f8446da76809cda3efb2fb : FAILURE in 27m 27s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ea29dac391bf697cf2ab99b8ab076829b7d1ce28","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-10-12 15:04:45.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/836252 : FAILURE in 21m 26s","accounts_in_message":[],"_revision_number":10},{"id":"4c7c543a4219dbe667d0aeef12d819980149acfc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-12 15:23:44.000000000","message":"Patch Set 9:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/727/","accounts_in_message":[],"_revision_number":9},{"id":"c15c05268bf596a1381d139fb404e331e03ddcc1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-10-12 15:30:46.000000000","message":"Patch Set 10:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c60e6011450f479db7939099385bbd30\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a4afa977018442deb4fd07a140b8aeac : SUCCESS in 39m 10s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cdcb0d0be61e496285d7e066c70bef46 : FAILURE in 44m 40s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ad17c42cd3584557a12016622d3d0ec6 : FAILURE in 1h 45m 01s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c0fae5de42ec4925a5ba3049b2349b71 : SUCCESS in 36m 24s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3af1526bb9d644f99d0fc281e9f6fe67 : RETRY_LIMIT in 35s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b25bad5669544d3aba1d8b9e1f1ba0b2 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6945fe3bbb9842e4a3e98f633d9c8916 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/622bbdb7ed854b6ca2765f4d3410e70a : RETRY_LIMIT in 26s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#479\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#436\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.ya... (truncated)","accounts_in_message":[],"_revision_number":10},{"id":"3e5464a6fe7e5f63e18f6d314ff036f459f4a4f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-12 15:58:25.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce9fb2322a9b4f6a97ac66e7cadb86d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d05bedc97b3b4b2c96c615c982293523 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8b8afc80bb648518f4d159a55bc9435 : SUCCESS in 10m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f7289a5c8114998af402372ea37378e : FAILURE in 9m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/977bcc6f23714eba921e34200534f4d5 : SUCCESS in 11m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/7591803d97fe4610bf0260f5283961a9 : SUCCESS in 37m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/62a220946079414abf0295e9e0dad06f : SUCCESS in 50m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f2291a38bcaa49e8ba0a7878219aae60 : SUCCESS in 1h 39m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92ec5beab29e422abe7396118734ff6a : SUCCESS in 1h 03m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c58499a2d7e48dfb79df8a58c4816db : SUCCESS in 15m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/44258fd6030f433493ed8745a56c47d2 : SUCCESS in 15m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/650b06cb7ca04a20825e48f4e155d034 : SUCCESS in 5m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c652dda60d554737b2ba886df485733f : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fcdf1fe74c7f49acaebf8fe885f664b1 : SUCCESS in 5m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/638b8f846ad14174bbeaba44b0626de8 : SUCCESS in 8m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e9989b7c89d048dca1ca68fe9bf63c66 : FAILURE in 29m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d8259bc2d40c4d7d8554b9f69e4a4309 : SUCCESS in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7a2530a6e37b4230b316fe065b4bfac8 : SUCCESS in 1h 34m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/756ee8c563e5428b8b7feebe5ca02daf : FAILURE in 2h 17m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b4a352c813874890b30ade8fa8e6ea4a : SUCCESS in 1h 46m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/49c3fd02f0244ff297742d315bed5598 : SUCCESS in 1h 43m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5c676feeda534f7998dbf8ca7775c5b2 : SUCCESS in 34m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3d2383dd48234a84a5a3ea13374acdfe : SUCCESS in 1h 07m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/11cc0a8c45914c86803c93565c317c95 : SUCCESS in 1h 22m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b08646fed43743e9bddbc3426cee46ee : SUCCESS in 41m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0cda492dd2b54a7a82853d840caeb8ba : SUCCESS in 1h 27m 54s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0bf143eda32648018e9eb75174ef1f6d : POST_FAILURE in 1h 28m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e90dc5c8a3e74662bea0fc5c114d9a97 : SUCCESS in 1h 06m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d410fbad9b146eca83251d13b7d25e9 : SUCCESS in 52m 46s","accounts_in_message":[],"_revision_number":10},{"id":"043ed6ed0bf143d545711ce52ca47be2bec10eee","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-12 16:00:16.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d2c041097a3e0674c839ad4ecc675249afe78dec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-12 16:22:25.000000000","message":"Patch Set 9:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/836252/9/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/836252/9/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"652a3773a44e31b04bd57934424153ac4a9e7478","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-12 16:22:42.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f74dfab95b35479bbc1d5aa05be8322e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/831257dab5014a389ec43411c803ab91 : SUCCESS in 17m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/841744d790684294ab0a8c592c7fb606 : FAILURE in 20m 46s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ca2558e2b59f0cbb31c7051f7aa9c004d2e4373c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-10-12 17:14:22.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/836252 : FAILURE in 21m 55s","accounts_in_message":[],"_revision_number":10},{"id":"148437f9ead9f53036f0c31adf4878e9a38aec4d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-12 17:21:34.000000000","message":"Patch Set 10:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/729/","accounts_in_message":[],"_revision_number":10},{"id":"0a903d9c1b55539a5513bda0b24ecb835d7503ed","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-10-12 18:05:55.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/836252 : SUCCESS in 52m 14s","accounts_in_message":[],"_revision_number":10},{"id":"a3e168c21995cfd1b1b8ba81a5da4326d3b6228f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-10-12 18:21:04.000000000","message":"Patch Set 10:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/836252/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/836252/10/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"a32a174954d6fd2b6cd8c5b2d06336b01d346b80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-12 18:36:16.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e45316fcaed349e1b39033131a783af6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3ac5d11e16945a79642ab654ecc9662 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40b7084ed1d345c594807a6ecaf4acbe : SUCCESS in 9m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9cdc2fb05a624ad5894978f930a3e70a : SUCCESS in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7d1dade8ab74576b6496964e8bc1531 : SUCCESS in 10m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/d268b0ea37734d699024197b3cd07b27 : SUCCESS in 55m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/07ff902ad5da4acfb65319578cee6754 : SUCCESS in 55m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6ca80e10f6af4a5898c401c1c125adf2 : SUCCESS in 1h 21m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8d187de53014bdab9d4d7864cba6002 : SUCCESS in 56m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd28c865173843d6893da3b42e1104c5 : SUCCESS in 18m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/635337513e1d4716842b54f241ba5744 : SUCCESS in 17m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3d6eac9795b7487498e9e2e0b7a1aa9f : SUCCESS in 3m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ab95a091d1f5434da4fce5edeea4c9c2 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dc2c33f860a24877bbe8835af609db80 : SUCCESS in 4m 37s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6dc4ca5cfd9a440daa11ebc005d2179d : SUCCESS in 5m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9b7d1cfc8a084bd880e111ae88f7475e : FAILURE in 17m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f7881c6f9d1c4689a8142d0d3831d00b : SUCCESS in 3m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6db403d0b7334936ae05c5ee424499f1 : SUCCESS in 2h 04m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e2daeae518724be392e10fe77f0929ba : FAILURE in 2h 22m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/45fe0e3f60994dcba7e19afefab66cb8 : FAILURE in 1h 55m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4bbff4782f0b4787bbd05ca44ea23f5d : SUCCESS in 1h 55m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/49151116dd8346cabeb1f8592b1566ca : SUCCESS in 42m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0c6768de1a20402a84b88a745c5ec38d : SUCCESS in 1h 10m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d9d4367e35014ddf9d1f56f8b4435109 : SUCCESS in 1h 50m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/810ccd36445c4f75adb6c6182332cd40 : SUCCESS in 51m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b21b792dabdb4715b1d2bf5d296f7c8c : SUCCESS in 1h 36m 11s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/3f7b4492e7764b4d831982bc7a45c00d : SUCCESS in 1h 05m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/31b85815713b45df9132c73915b4f505 : SUCCESS in 1h 11m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/671419484e4f44cdba9954516b8644c8 : SUCCESS in 39m 07s","accounts_in_message":[],"_revision_number":10},{"id":"b8528a61e99940ca640451b89cfb0e8f2bb1c8bf","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-12 19:13:24.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4b4025c20147449d9c02fd9e31a577c0 : FAILURE in 1h 12m 42s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a17e86840e1343bd87644447240c7823 : FAILURE in 1h 21m 17s","accounts_in_message":[],"_revision_number":10},{"id":"75f204f38f2d332a283bb7fc8cadb4389f5b21b9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-13 01:57:03.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"73b19d1c47d8f7916bd77cf4f14a0c7707eee0d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-13 02:23:08.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21eedb0bfdf84b40bc7bb0307b1f1343\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1337a887e4364953aa39be834c3b399f : SUCCESS in 19m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d19e07f59c0b4a69b0a0b276a02baa4c : SUCCESS in 23m 56s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e337ec5e8746e06a01361f6964915a4ce40070fd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-13 03:33:42.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4b0edf77f3f84430bf97cedeac9f0a8e : FAILURE in 1h 15m 42s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/47738b66e7304e1eafe0faceae9626ad : FAILURE in 1h 34m 33s","accounts_in_message":[],"_revision_number":10},{"id":"fa687dd172acc9f9dbcb4607b291f8f569c8f068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-13 04:16:01.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b5228a9eb724512a64d1ae64a75e722\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b90472b0e414f1e955c94a726e237ba : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe1d4e3b7a84429aa6dd4fba411b9303 : SUCCESS in 10m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a346a6398b234e08b7cade6fa8e08455 : FAILURE in 9m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50e1c2649684bc0b119a6ad4e7e0f5a : SUCCESS in 12m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/06e156fae3994bb2a5005d8682668774 : SUCCESS in 38m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/51a894e6107d422c9359f4a60c75bcd4 : SUCCESS in 1h 00m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/62cd89ac4ea44202a2430b431dc253e3 : SUCCESS in 1h 33m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1eb1625d51154e93b2674e01a5aafc9e : SUCCESS in 57m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/845cc8958a534304b52dfb7033947dee : SUCCESS in 14m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4dc20aae4e8f4d498d8e9cac9ca667ae : SUCCESS in 18m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1214e12703274bfc92211e2e57015301 : SUCCESS in 4m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a78ee5ca76bc40e3adcda52314650ce7 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d8b15409dea34e3a910ce4b88edf51cb : SUCCESS in 6m 02s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2f41b9f6bdc24c128833c4d45e750fe4 : SUCCESS in 5m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/86a2de82f4a242a58455fad7b9756410 : FAILURE in 22m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ee8a98de05b3411baf5576697df373f3 : SUCCESS in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a781a84813b5483fb1cd608693d740a4 : SUCCESS in 1h 51m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/74c912c975a94055bc559f53b38085b1 : FAILURE in 2h 17m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cab12bfc268c4728b18f9cbf7b66028c : SUCCESS in 2h 03m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a85be2d76abb4b73919c0667d2aaf5ac : SUCCESS in 2h 05m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9f6833a3b8c14d429c7777af8f5afff4 : SUCCESS in 22m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/23fe60ea45cf42bfafe17d700d34f2e2 : SUCCESS in 1h 16m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1a1aef6e6a3e47aa9e67898960ccdaed : SUCCESS in 1h 47m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/79fe82dc8e4547eb8826bf3166eef0e0 : SUCCESS in 30m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e1e830f78e604d53a8e3954ef9e7b0c9 : SUCCESS in 1h 23m 43s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2f05928c58784d6d86dac9d8d4ef6561 : SUCCESS in 1h 05m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c7782c8ba066489fb664c6e585fc3470 : SUCCESS in 1h 49m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf83a662ff674a60874497084925ceff : SUCCESS in 51m 07s","accounts_in_message":[],"_revision_number":10},{"id":"8ba35ed38911cc5840be0ff0e9ee733080dfef95","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-13 13:03:48.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/52/836252/10/check/DellEMC_PowerFlex_3.6/7c8eb04/DellEMC_PowerFlex_3.6/2231 : SUCCESS in 2h 31m 51s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/52/836252/10/check/DellEMC_PowerFlex_4.x/e4808bc/DellEMC_PowerFlex_4.x/1113 : SUCCESS in 2h 05m 07s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/52/836252/10/check/DellEMC_PowerFlex_4.x_NVMeTCP/ad3c762/DellEMC_PowerFlex_4.x_NVMeTCP/1039 : FAILURE in 5h 55m 00s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"aa7f127a6cf84314ce2cbdff657e4343ae3a3e67","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-10-13 22:30:08.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/52/836252/10/20241014014606/NEC-ISCSI : SUCCESS in 5h 35m 56s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/52/836252/10/20241014014606/NEC-FC : SUCCESS in 5h 43m 58s","accounts_in_message":[],"_revision_number":10},{"id":"0972f8510c44c80d558f7c2956af0263ac4e6098","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-10-13 23:40:10.000000000","message":"Patch Set 10:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/52/836252/10/20241013230143/NEC-V-ISCSI : SUCCESS in 3h 58m 22s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/52/836252/10/20241013230143/NEC-V-FC : FAILURE in 9h 38m 23s","accounts_in_message":[],"_revision_number":10},{"id":"e2e18b7c6c910fbf3f270341e0a8e12209c910c7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-14 04:49:24.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/52/836252/10/check/DellEMC_Unity_FC/3c72393/DellEMC_Unity_FC/2732 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/52/836252/10/check/DellEMC_Unity_iSCSI/be7f10d/DellEMC_Unity_iSCSI/2729 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"15e5d4e55f4b88ea1c22e735fcafd0d6bd9e66fd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-14 05:38:30.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"aa1c793cb2a615741f8cbf87cbc265d05dfce13c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 06:04:14.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85f2e151d1c645d39c4a960843f274e7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29f04fd3e4ab461ab357dbe29e70c98f : FAILURE in 23m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b1ea0528664242ebad97af25080e8239 : FAILURE in 23m 37s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"38e5d0ba6cc06af88618a4a707a39f30f5d950e3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-10-14 06:58:25.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ee37bbcf3a17473e861bb1f2f560079f : SUCCESS in 1h 13m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e1308f9c4f094f53b729a875b79de6d8 : SUCCESS in 1h 17m 52s","accounts_in_message":[],"_revision_number":10},{"id":"5bed38b3f48db842337dc7597094580838029e96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 07:53:55.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b59f6c77029049768b2b3af0135333bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4f4b04dbaac4aee87cccbbaa03e3a65 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf45cca083064cd788eeaade72f903b5 : SUCCESS in 11m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8274e0da7fb04a59a326a37b040f174f : SUCCESS in 11m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c9081766b4f48e191d18c9542bbb2d4 : SUCCESS in 11m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/c8b2e54ab0a94f2d8913a70c7ed3a831 : SUCCESS in 44m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e8be07b0b57447c7aa7053470d93e882 : SUCCESS in 43m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0defa290e7624a209fbe336ab0537f95 : SUCCESS in 1h 12m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48967a6211944c76930ae1e5f3f8a22d : SUCCESS in 47m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f249e21393ee4c48bda396231df6f903 : SUCCESS in 15m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9804c1658e56409eb835b0ee3073ad44 : SUCCESS in 14m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3b6da2e68d214c4e89c14f903962f221 : SUCCESS in 5m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0f66c804631b4581990d61ae3e70ba3b : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f757f217f12840a3a1ef17983b8391e0 : SUCCESS in 6m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/271078f54b4c44089cc72c68b0269305 : SUCCESS in 5m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c024870200a248ddaf802cee6fac8f51 : FAILURE in 32m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0e5f5204ebe5414c9d9608f1f2936bbc : SUCCESS in 4m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/66184997302b40dda34aa5e116207459 : SUCCESS in 1h 54m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/928a72a5929b484594b3fcb827cb3550 : FAILURE in 1h 33m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8116a46529d34ea6a5badeaf619a2bbd : SUCCESS in 1h 45m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2e7c05a4a25d4900b7befc8e266160b3 : SUCCESS in 2h 08m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2a54bd6751c841dda947cad9cdf2a080 : SUCCESS in 22m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6bf62a3edf104f3c91e3a5c1edd84283 : SUCCESS in 1h 11m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e7189182a4574d549dad8235f28e5fd0 : SUCCESS in 1h 48m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/503906ecd57944338a9a5a903d6a9708 : SUCCESS in 24m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7a34dda140d24c23a29a0a6cc05c4da4 : SUCCESS in 1h 01m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/47f79dde75de4eccb3aee73b70462687 : SUCCESS in 1h 10m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d33443a99095403c821be6dd448de0f9 : SUCCESS in 1h 07m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/346d68dc264e41b2b864121f52e0f5a6 : SUCCESS in 37m 14s","accounts_in_message":[],"_revision_number":10},{"id":"ee04d8bfb7bb288df51b50231f71a89285604aae","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-10-14 07:59:44.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_10_2024_07_54_05\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"bc402ac8395494ba46a4a359bb73220486fd48dc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-10-14 08:10:30.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/52/836252/10/upstream-check/cinder-tempest-plugin-ontap-iscsi/74ce13d/ : RETRY_LIMIT in 5m 07s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/52/836252/10/upstream-check/cinder-tempest-plugin-ontap-nfs/188a9a6/ : RETRY_LIMIT in 4m 48s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"5050e69be7200edd283c149fe60c1f59d7a7bb62","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-10-14 10:54:16.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/52/836252/10/check/DellEMC_PowerStore_iSCSI/c9b0c89/DellEMC_PowerStore_iSCSI/1817 : SUCCESS in 1h 32m 34s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/52/836252/10/check/DellEMC_PowerStore_FC/7f9c7b8/DellEMC_PowerStore_FC/2015 : ABORTED in 3h 00m 01s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/52/836252/10/check/DellEMC_PowerStore_NFS/9e68cd7/DellEMC_PowerStore_NFS/1510 : SUCCESS in 1h 08m 11s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/52/836252/10/check/DellEMC_PowerStore_NVMeTCP/5979cbd/DellEMC_PowerStore_NVMeTCP/1369 : SUCCESS in 2h 05m 13s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/52/836252/10/check/DellEMC_PowerStore_NVMeFC/8b9e18d/DellEMC_PowerStore_NVMeFC/334 : SUCCESS in 1h 15m 55s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"403afbbb4d31acf500e96cab1c43726312fc0494","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-10-14 13:47:46.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/52/836252/10/check/DellEMC_PowerFlex_3.6/d1b61ea/DellEMC_PowerFlex_3.6/2235 : SUCCESS in 2h 26m 17s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/52/836252/10/check/DellEMC_PowerFlex_4.x/9f4bc00/DellEMC_PowerFlex_4.x/1117 : SUCCESS in 2h 02m 12s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/52/836252/10/check/DellEMC_PowerFlex_4.x_NVMeTCP/e48ec49/DellEMC_PowerFlex_4.x_NVMeTCP/1043 : FAILURE in 5h 53m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"ecf72310405ae09add8736060670f1fc2a9bb0eb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-10-15 04:49:52.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/52/836252/10/check/DellEMC_Unity_FC/68b7e18/DellEMC_Unity_FC/2736 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/52/836252/10/check/DellEMC_Unity_iSCSI/6bb8c45/DellEMC_Unity_iSCSI/2733 : ABORTED in 6h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"ba474b9d08d7b4042acd74656c35e593a7dc0ea5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-20 15:41:11.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"b5011fc601d630d4c94463c506ab826753fb87cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-20 16:10:43.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41f74435d82545eaa0d5cd82762ec263\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a295626123d34400adbe58893e0dc9de : SUCCESS in 22m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cfdd4765a36c46968cd2fa4be3958824 : SUCCESS in 21m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ec52e6b2adec46558b183fe48ef40639 : SUCCESS in 27m 10s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"8330cfcf7df2f512d28ab83d30d7a5ff27405167","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-20 16:27:37.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/836252 : SUCCESS in 44m 44s","accounts_in_message":[],"_revision_number":11},{"id":"660ecaf937dfdb5613bb885dc52b538cebbac590","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-20 16:36:22.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1279/.","accounts_in_message":[],"_revision_number":11},{"id":"4a159c4aa2e68dba23c69b406f892f41a9c80bcd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-20 16:37:00.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/886ade2183d14079a3599dde7ffbf667 : FAILURE in 53m 00s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d581345ecaae47a7a2c43d1689d28c35 : SUCCESS in 46m 39s","accounts_in_message":[],"_revision_number":11},{"id":"595b753b5a75553fd46beb51286e1d00c3a0b883","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-20 16:41:15.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1412/.","accounts_in_message":[],"_revision_number":11},{"id":"9898b1b1dc6322c821e13e697254092e9312971c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-20 16:41:51.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/836252 : SUCCESS in 59m 20s","accounts_in_message":[],"_revision_number":11},{"id":"1016311bdaf40484c457a9c45c95ebd626a7725c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-20 16:51:51.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/69e84d864b9b455eb6fb1c129db10060\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4fd185f2003640919f1ddee78cf6c8f1 : NODE_FAILURE Node request 200-0000075608 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/acf0be62189c4de2bd24b8414b18c268 : NODE_FAILURE Node request 200-0000075609 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/818ff409a71249fba80640dcd84cd3d5 : SUCCESS in 57m 26s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3618b0b23d1e4b24b7be4fa26395ebfa : FAILURE in 1h 10m 04s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c463f9d365d043b0872f0a5a34a7abe3 : SUCCESS in 58m 11s","accounts_in_message":[],"_revision_number":11},{"id":"077a72f4bb6e3bc4d2aa774e1a6a6c7a22ba4f91","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-20 17:41:42.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n\\* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"243660bed630a155cfcf3a043fb95b9a007c1fca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-20 18:00:25.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/025a5ad5866843d2838acb1da1d5d115\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5acc281481174b68926a46f719611fa2 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c500539f274f44a19e8022490782a599 : SUCCESS in 11m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7141483435ea4574bb36f9f9cf875a83 : SUCCESS in 10m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01221e7100ed40bda0357b5188a6c8a1 : SUCCESS in 14m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27e3fcf30ec84af697593f841a14c6c7 : SUCCESS in 13m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/aa471488b04e48139d9be179619d552f : SUCCESS in 1h 03m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aebbd651cd7c4bf18db02d7f854b47f1 : SUCCESS in 1h 01m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3fff2e8fb0074718ab2361c533180017 : SUCCESS in 1h 35m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31e1ee2fa2c146c6ad91c1b93d3371f5 : SUCCESS in 1h 03m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28d2cb3c4d624d619bf6666ab75cae19 : SUCCESS in 13m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e60cd81a67184d7da0d86a263e0dc0d7 : TIMED_OUT in 40m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/086db4ab25c04ccb9744e6112b9283aa : SUCCESS in 4m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/84c7564726324788be815793564ae44d : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3c7af4d259b74265bb6053d5deac9bcc : SUCCESS in 3m 33s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c003c766a9af4156b004e0e0b54c691f : SUCCESS in 4m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3ee03519c72641c291eea919080a59c9 : FAILURE in 37m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8a3f8eb77dd841f8b4c0e76cd99ef7a9 : SUCCESS in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8629574aadd64058800429fe846e440a : SUCCESS in 1h 46m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/186cbcb77db346118aaf32d4ace9516f : FAILURE in 2h 13m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/400850e14ae54b45837a91581c0766fd : SUCCESS in 1h 37m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/dc744185618d44cc99985b8b1660a664 : FAILURE in 13m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/303e85499cb940519086628c57c020cf : SUCCESS in 24m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/16d2fbf0b4c543f1a68b537e980085eb : SUCCESS in 1h 03m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f601a0be4bd4443db202e796e372ceca : SUCCESS in 1h 29m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2650add4e9574b3fb4417cd2e0b6d9d9 : SUCCESS in 39m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7065b5c3344c4c5e80276160bb7661bb : SUCCESS in 1h 18m 16s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d0ac7d9c78014273bdfeb89cf3aa84b0 : FAILURE in 13m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f2365e33e14843fb97629867993d2741 : SUCCESS in 1h 27m 40s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99aad23c5d744fc3a4d770f900dcd2a9 : SUCCESS in 1h 24m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cac979b1d4340ea8a943c688647a555 : SUCCESS in 51m 09s","accounts_in_message":[],"_revision_number":11},{"id":"80f1b9d0e81b8f8f53e2cb2edecf826a67974402","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-20 18:34:07.000000000","message":"Patch Set 11:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-52-836252-11/FC/ : FAILURE in 2h 38m 16s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-52-836252-11/iSCSI/ : FAILURE in 2h 52m 26s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":11},{"id":"99bb6286b4d2075db358b86515f0bd322b64cac0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-24 14:16:05.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"e1b4d98ffe0be57eb8d654b9ebb2f990ef1dde35","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-25 06:27:18.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/52/836252/11/20250725095405/NEC-V-FC : SUCCESS in 5h 29m 15s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/52/836252/11/20250725095405/NEC-V-ISCSI : SUCCESS in 5h 33m 10s","accounts_in_message":[],"_revision_number":11},{"id":"e7e8237730772b6e8154a137253b615892cceb61","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-07-25 10:25:38.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/52/836252/11/20250725095346/NEC-ISCSI : SUCCESS in 5h 51m 00s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/52/836252/11/20250725095346/NEC-FC : SUCCESS in 9h 31m 49s","accounts_in_message":[],"_revision_number":11},{"id":"49358e0cdba0585094dec9120057c736b510accc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-11 13:48:07.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"259e0b7da3caaba280470c23035e473227eefc69","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-05-11 14:22:31.000000000","message":"Patch Set 12:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/836252 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"5e794019a6c4e04d79f3eb97dfc2dc1ec40dc6f0","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-11 14:24:45.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-fc-driver build/f236e3aa855f4a93b64c1f3082df66f2 : FAILURE in 36m 13s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/836252/etc/tempest_log.log","accounts_in_message":[],"_revision_number":12},{"id":"75766222d59e5f549af72949e43d18e4406687f9","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-11 14:32:34.000000000","message":"Patch Set 12:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/836252 : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"4adc94c0cdb2d1936a4d6aeea6a1986dda91d2ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 14:52:18.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/522265875aa84c80922182c5fa6a351a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6f3e1e10adb749c39a25d4af62e9a7c5 : FAILURE in 5m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8386838fd59744c88905c5aae05855c7 : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c67530c8d0b04deebcf6866443e8afcc : TIMED_OUT in 1h 01m 33s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"aa4f9d313bacae667b9e0490df471e27850d7a65","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-11 14:52:21.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0dd59ed5d7e63ea7a5e80c6a0505e6e1f1f60045 .","accounts_in_message":[],"_revision_number":12},{"id":"745abdb8eaac0e84b0ee140d90b544e257b44b9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 16:26:34.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48462140036b40feaf361ebdf392adbc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dc210d62bdd48838caaf3a091d3da3c : SUCCESS in 7m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59b603607688452894fba63aee46dd49 : SUCCESS in 12m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6174bfdd8ecc49248bb1aa46dc2a4dbb : SUCCESS in 8m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/01c23f11df41488baef68c9dd86d366a : SUCCESS in 15m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2b6b8707b984b3080940a77db039087 : SUCCESS in 12m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/b7d1d454a4c845ce89ad9f3b5268fcb8 : SUCCESS in 1h 01m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d4f77218cc0240b287046f660eaefae6 : SUCCESS in 1h 11m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/56864343216d43998de36705a927fb12 : SUCCESS in 1h 42m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a49d53f8993442b286ec96ada1353675 : SUCCESS in 1h 05m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6e0a8211e9445fd8f6bfe80d1188d78 : SUCCESS in 23m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9fd2542bed004ef08fe38dcc8937ba49 : SUCCESS in 28m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e1edd387a76e4b62a6f5b09621c81f52 : SUCCESS in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b2afcd33f1fc4b9080169f1f8cabfc06 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a91e53c8f51d44f392310eb8df876016 : SUCCESS in 6m 20s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cb6662d808e849a9b1e8e4c213f8ab33 : SUCCESS in 5m 25s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/af63294923f845cc862cb697e6fbef76 : SUCCESS in 9m 26s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/034367bbcf5e40cb9b45a0c28fa38bf7 : SUCCESS in 32m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a50fa089bcb949e587f13a1377810b30 : FAILURE in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2b4ccae8ec064a4c88b564ab85cd23dc : SUCCESS in 1h 05m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/481c39073f1946649dda5dfbb67bdda6 : FAILURE in 2h 37m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/69a29c84ebe24aa9a83ae8ad05b052b4 : SUCCESS in 2h 07m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a54661a5e5a643db82f36b51e087d5bc : SUCCESS in 2h 07m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f90a48ed4c31465595bc907c5ac5eff7 : SUCCESS in 36m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b215bd5ba0684a00a41b2d6c5c473df8 : SUCCESS in 1h 17m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ad53826b6abd4d7d9a14e074ce493a68 : SUCCESS in 1h 44m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/357bc61e9093464b838c84ee421606e5 : SUCCESS in 41m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/68cab19fc31448fb919360768c40c462 : SUCCESS in 1h 50m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8eccbe3a1434988b5c5f5b9f763e34d : SUCCESS in 56m 23s","accounts_in_message":[],"_revision_number":12},{"id":"e8ce419d28fa605359c04175dbb993f8a3b79695","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-16 13:17:23.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/4416582a49a3430293366d2e5ddd5115 : FAILURE in 29m 07s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/836252/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/836252/etc/tempest_log.log","accounts_in_message":[],"_revision_number":12},{"id":"e2e4fd5c22ca9bde55cdd2ace0d0f20371ac52f7","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-05-26 11:12:09.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"76ecb1cbc95d1873282d984690bae3af6729be0d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:02:14.000000000","message":"Patch Set 12:\n\nBuild Failed","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"8f999aa0dfeca3ee557de84de3b42f6a7d594654","revisions":{"e69ea6c5eabd96b19fcb37bcadd9ed0ed06239c7":{"kind":"REWORK","_number":1,"created":"2022-04-03 15:37:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/836252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/52/836252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/52/836252/1"}}},"commit":{"parents":[{"commit":"3d3b06179a4d955665fb167f7fd25947a0779e2a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d3b06179a4d955665fb167f7fd25947a0779e2a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 15:36:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 15:37:13.000000000","tz":540},"subject":"Deprecate [DEFAULT] use_forwarded_for","message":"Deprecate [DEFAULT] use_forwarded_for\n\n... because functionality of this parameter is effectively duplicate of\nthe HTTPProxyToWSGI middleware in oslo.middleware library.\n\nChange-Id: I0dd59ed5d7e63ea7a5e80c6a0505e6e1f1f60045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e69ea6c5eabd96b19fcb37bcadd9ed0ed06239c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e69ea6c5eabd96b19fcb37bcadd9ed0ed06239c7"}]},"branch":"refs/heads/master"},"5ae29da23baa958fe65e901001e657f92e70b9dd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-04-03 15:49:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/836252/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/52/836252/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/52/836252/2"}}},"commit":{"parents":[{"commit":"3d3b06179a4d955665fb167f7fd25947a0779e2a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d3b06179a4d955665fb167f7fd25947a0779e2a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 15:36:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 15:49:12.000000000","tz":540},"subject":"Deprecate [DEFAULT] use_forwarded_for","message":"Deprecate [DEFAULT] use_forwarded_for\n\n... because functionality of this parameter is effectively duplicate of\nthe HTTPProxyToWSGI middleware in oslo.middleware library.\n\nCloses-Bug: #1967686\nChange-Id: I0dd59ed5d7e63ea7a5e80c6a0505e6e1f1f60045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ae29da23baa958fe65e901001e657f92e70b9dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ae29da23baa958fe65e901001e657f92e70b9dd"}]},"branch":"refs/heads/master"},"ad2cfa3d86e9b003d5a13686e0cdcd9074cb8dbf":{"kind":"REWORK","_number":3,"created":"2022-04-11 23:16:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/836252/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/52/836252/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/52/836252/3"}}},"commit":{"parents":[{"commit":"3d3b06179a4d955665fb167f7fd25947a0779e2a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d3b06179a4d955665fb167f7fd25947a0779e2a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 15:36:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-11 23:16:03.000000000","tz":540},"subject":"Deprecate [DEFAULT] use_forwarded_for","message":"Deprecate [DEFAULT] use_forwarded_for\n\n... because functionality of this parameter is effectively duplicate of\nthe HTTPProxyToWSGI middleware in oslo.middleware library.\n\nCloses-Bug: #1967686\nChange-Id: I0dd59ed5d7e63ea7a5e80c6a0505e6e1f1f60045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad2cfa3d86e9b003d5a13686e0cdcd9074cb8dbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad2cfa3d86e9b003d5a13686e0cdcd9074cb8dbf"}]},"branch":"refs/heads/master"},"56718817622b53bf0ed4d812288f036d458e5d20":{"kind":"REWORK","_number":4,"created":"2022-07-16 14:13:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/836252/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/52/836252/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/52/836252/4"}}},"commit":{"parents":[{"commit":"3d3b06179a4d955665fb167f7fd25947a0779e2a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d3b06179a4d955665fb167f7fd25947a0779e2a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 15:36:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-16 14:13:58.000000000","tz":0},"subject":"Deprecate [DEFAULT] use_forwarded_for","message":"Deprecate [DEFAULT] use_forwarded_for\n\n... because functionality of this parameter is effectively duplicate of\nthe HTTPProxyToWSGI middleware in oslo.middleware library.\n\nCloses-Bug: #1967686\nChange-Id: I0dd59ed5d7e63ea7a5e80c6a0505e6e1f1f60045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56718817622b53bf0ed4d812288f036d458e5d20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56718817622b53bf0ed4d812288f036d458e5d20"}]},"branch":"refs/heads/master"},"09af02f3c47d1fc978aa669b1b51f50eba4a95b5":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-07-16 14:14:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/836252/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/52/836252/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/52/836252/5"}}},"commit":{"parents":[{"commit":"3d3b06179a4d955665fb167f7fd25947a0779e2a","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d3b06179a4d955665fb167f7fd25947a0779e2a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 15:36:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-16 14:14:20.000000000","tz":0},"subject":"Deprecate [DEFAULT] use_forwarded_for","message":"Deprecate [DEFAULT] use_forwarded_for\n\n... because functionality of this parameter is effectively a duplicate\nof the HTTPProxyToWSGI middleware in oslo.middleware library.\n\nCloses-Bug: #1967686\nChange-Id: I0dd59ed5d7e63ea7a5e80c6a0505e6e1f1f60045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/09af02f3c47d1fc978aa669b1b51f50eba4a95b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/09af02f3c47d1fc978aa669b1b51f50eba4a95b5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a6154395fc29d0edf7f05e681e6e81fd3728110f":{"kind":"REWORK","_number":6,"created":"2023-07-22 14:37:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/836252/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/52/836252/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/52/836252/6"}}},"commit":{"parents":[{"commit":"d69e26244e74d65ad9597aeb9f74b781859aa651","subject":"Merge \"mypy: Cover cinder/flow_utils.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d69e26244e74d65ad9597aeb9f74b781859aa651"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 15:36:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-22 14:37:32.000000000","tz":540},"subject":"Deprecate [DEFAULT] use_forwarded_for","message":"Deprecate [DEFAULT] use_forwarded_for\n\n... because functionality of this parameter is effectively a duplicate\nof the HTTPProxyToWSGI middleware in oslo.middleware library.\n\nCloses-Bug: #1967686\nChange-Id: I0dd59ed5d7e63ea7a5e80c6a0505e6e1f1f60045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a6154395fc29d0edf7f05e681e6e81fd3728110f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a6154395fc29d0edf7f05e681e6e81fd3728110f"}]},"branch":"refs/heads/master"},"e4f4eca0f8af7cc88058144a0e18a2da9aaf993e":{"kind":"REWORK","_number":7,"created":"2023-12-11 12:58:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/836252/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/52/836252/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/52/836252/7"}}},"commit":{"parents":[{"commit":"d69e26244e74d65ad9597aeb9f74b781859aa651","subject":"Merge \"mypy: Cover cinder/flow_utils.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d69e26244e74d65ad9597aeb9f74b781859aa651"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 15:36:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-11 12:58:12.000000000","tz":0},"subject":"Deprecate [DEFAULT] use_forwarded_for","message":"Deprecate [DEFAULT] use_forwarded_for\n\n... because functionality of this parameter is effectively a duplicate\nof the HTTPProxyToWSGI middleware in oslo.middleware library.\n\nCloses-Bug: #1967686\nChange-Id: I0dd59ed5d7e63ea7a5e80c6a0505e6e1f1f60045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e4f4eca0f8af7cc88058144a0e18a2da9aaf993e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e4f4eca0f8af7cc88058144a0e18a2da9aaf993e"}]},"branch":"refs/heads/master"},"48f0c45130f4acd236288c0bc855ca9f706f904c":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-12-11 12:58:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/836252/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/52/836252/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/52/836252/8"}}},"commit":{"parents":[{"commit":"e9f94157895516c1c75052f6c982d9cb4aaeaa2b","subject":"Merge \"RBD: tpool.Proxy client object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9f94157895516c1c75052f6c982d9cb4aaeaa2b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 15:36:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-11 12:58:19.000000000","tz":0},"subject":"Deprecate [DEFAULT] use_forwarded_for","message":"Deprecate [DEFAULT] use_forwarded_for\n\n... because functionality of this parameter is effectively a duplicate\nof the HTTPProxyToWSGI middleware in oslo.middleware library.\n\nCloses-Bug: #1967686\nChange-Id: I0dd59ed5d7e63ea7a5e80c6a0505e6e1f1f60045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/48f0c45130f4acd236288c0bc855ca9f706f904c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/48f0c45130f4acd236288c0bc855ca9f706f904c"}]},"branch":"refs/heads/master","description":"Rebase"},"2af66da4b0f04a7fb2b421c219e81d6d552f483e":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-10-12 13:28:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/836252/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/52/836252/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/52/836252/9"}}},"commit":{"parents":[{"commit":"774d784aca50c942545fde5512f7de73b1caa5be","subject":"Merge \"Tests: Clean up RBD image unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/774d784aca50c942545fde5512f7de73b1caa5be"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 15:36:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-12 13:28:09.000000000","tz":0},"subject":"Deprecate [DEFAULT] use_forwarded_for","message":"Deprecate [DEFAULT] use_forwarded_for\n\n... because functionality of this parameter is effectively a duplicate\nof the HTTPProxyToWSGI middleware in oslo.middleware library.\n\nCloses-Bug: #1967686\nChange-Id: I0dd59ed5d7e63ea7a5e80c6a0505e6e1f1f60045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2af66da4b0f04a7fb2b421c219e81d6d552f483e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2af66da4b0f04a7fb2b421c219e81d6d552f483e"}]},"branch":"refs/heads/master","description":"Rebase"},"d2a5c32e5a054102b10b155016ed3fda2101f093":{"kind":"REWORK","_number":10,"created":"2024-10-12 13:35:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/836252/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/52/836252/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/52/836252/10"}}},"commit":{"parents":[{"commit":"774d784aca50c942545fde5512f7de73b1caa5be","subject":"Merge \"Tests: Clean up RBD image unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/774d784aca50c942545fde5512f7de73b1caa5be"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 15:36:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-12 13:35:01.000000000","tz":0},"subject":"Deprecate [DEFAULT] use_forwarded_for","message":"Deprecate [DEFAULT] use_forwarded_for\n\n... because functionality of this parameter is effectively a duplicate\nof the HTTPProxyToWSGI middleware in oslo.middleware library.\n\nCloses-Bug: #1967686\nChange-Id: I0dd59ed5d7e63ea7a5e80c6a0505e6e1f1f60045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d2a5c32e5a054102b10b155016ed3fda2101f093"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d2a5c32e5a054102b10b155016ed3fda2101f093"}]},"branch":"refs/heads/master"},"70c578e2613831168e732e494b9cfa5d81f9a4ce":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":11,"created":"2025-07-20 15:41:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/836252/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/52/836252/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/52/836252/11"}}},"commit":{"parents":[{"commit":"21af57108f9f67ce9e5af78a6b7eef0d3f828576","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/21af57108f9f67ce9e5af78a6b7eef0d3f828576"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 15:36:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-20 15:41:01.000000000","tz":540},"subject":"Deprecate [DEFAULT] use_forwarded_for","message":"Deprecate [DEFAULT] use_forwarded_for\n\n... because functionality of this parameter is effectively a duplicate\nof the HTTPProxyToWSGI middleware in oslo.middleware library.\n\nCloses-Bug: #1967686\nChange-Id: I0dd59ed5d7e63ea7a5e80c6a0505e6e1f1f60045\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70c578e2613831168e732e494b9cfa5d81f9a4ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70c578e2613831168e732e494b9cfa5d81f9a4ce"}]},"branch":"refs/heads/master"},"8f999aa0dfeca3ee557de84de3b42f6a7d594654":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-05-11 13:48:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/836252/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/52/836252/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/836252/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/52/836252/12"}}},"commit":{"parents":[{"commit":"11c1a009150b803b115fb13fd9ab2a684bd9ea15","subject":"Merge \"Limit changes triggering the nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11c1a009150b803b115fb13fd9ab2a684bd9ea15"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 15:36:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-11 13:48:07.000000000","tz":0},"subject":"Deprecate [DEFAULT] use_forwarded_for","message":"Deprecate [DEFAULT] use_forwarded_for\n\n... because functionality of this parameter is effectively a duplicate\nof the HTTPProxyToWSGI middleware in oslo.middleware library.\n\nCloses-Bug: #1967686\nChange-Id: I0dd59ed5d7e63ea7a5e80c6a0505e6e1f1f60045\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f999aa0dfeca3ee557de84de3b42f6a7d594654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f999aa0dfeca3ee557de84de3b42f6a7d594654"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"70c578e2613831168e732e494b9cfa5d81f9a4ce","theirs":"11c1a009150b803b115fb13fd9ab2a684bd9ea15","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
