)]}'
{"id":"openstack%2Fcinder~836753","triplet_id":"openstack%2Fcinder~stable%2Fvictoria~I5108fd51effa4d72581654ed450d191a13e0e964","project":"openstack/cinder","branch":"stable/victoria","topic":"783389-stable/victoria","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2022-04-20 17:18:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5108fd51effa4d72581654ed450d191a13e0e964","subject":"Fix: Schema validation for attachment create API","status":"MERGED","created":"2022-04-06 09:12:32.000000000","updated":"2022-04-20 17:22:08.000000000","submitted":"2022-04-20 17:18:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"836753-783389-stable/victoria","meta_rev_id":"42923f049be453a3d257c872eb074c46110d869e","_number":836753,"virtual_id_number":836753,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-20 17:18:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"all":[{"value":1,"date":"2022-04-20 13:49:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":2,"date":"2022-04-20 14:54:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2022-04-08 05:14:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2022-04-20 14:54:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-06 09:12:32.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2022-04-06 10:05:30.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-06 10:07:13.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-04-06 10:33:47.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-04-06 10:43:57.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-04-06 11:49:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-06 11:50:56.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-04-06 11:54:09.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-06 11:55:00.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-06 12:06:03.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-04-06 12:27:36.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-06 14:56:15.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-08 05:14:27.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-04-20 13:49:48.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2022-04-20 14:54:37.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"e36d1620bdeb86498233d9d8075220d8d7c585ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2022-04-06 09:12:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4dab74d603d215ed648bb08aabcf8234496d59eb","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2022-04-06 09:16:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8af30d006077aa201bd055055b9c1c029abe97cf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-06 10:05:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/836753 : FAILURE in 51m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f65b4af58e0c57c7ba0712280ba18f7fe21c1668","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-06 10:07:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/53/836753/1/check/iscsi-driver-alletra-6k/a9903fc/a9903fc.html : FAILURE in 53m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5f467808fd9f816bfae45186ad1328163b4b311d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-06 10:08:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/836753 : FAILURE in 54m 01s","accounts_in_message":[],"_revision_number":1},{"id":"802f9c6f675cef707ca9a97386ed52905d1fbce4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-06 10:28:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/53/836753/1/check/iscsi-driver-nimble/5b35459/5b35459.html : FAILURE in 48m 25s","accounts_in_message":[],"_revision_number":1},{"id":"add4a58b17f23060faba568f9790119922631256","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-06 10:33:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/836753/1/iscsi/ : SUCCESS in 1h 20m 22s\n- cinder-smb http://cloudbase-ci.com/cinder/836753/1/smb/ : SUCCESS in 1h 07m 46s","accounts_in_message":[],"_revision_number":1},{"id":"30b0c167755f2f36cdb685b34600edc2b9188d43","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-06 10:43:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/53/836753/1/check/dsvm-tempest-cisco-zonemanager-job/c69e8ea/ : SUCCESS in 1h 20m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e2f45ab782ab12da98d419c83c1bd259c5681af5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-06 11:19:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/53/836753/1/check/iscsi-driver-nimble/45aea8b/45aea8b.html : FAILURE in 49m 34s","accounts_in_message":[],"_revision_number":1},{"id":"86b45b1ecc3ee5bf983cfc0b89a4ab395d36d6ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 11:49:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fc277196c23494398e0f109861c0ccf : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fa585248c274042931d1b52ddcfeae4 : SUCCESS in 9m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99af07233c514462bbad654ddd506b44 : SUCCESS in 9m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf096888ed794b12a258e3156e23d748 : SUCCESS in 14m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/baa324cc35e6447aba27f7ed1ebe63f2 : SUCCESS in 55m 57s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/39bbb6b5e2084e77814f32241062877a : FAILURE in 50m 25s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/60cb73e1aab44c83b0a01a02dd7b2f7c : SUCCESS in 1h 11m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73ae7d6dc1744ebeb8bd4f3ce8bb0c7a : SUCCESS in 51m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f79e09f329db40dcb525138f9a7106b8 : SUCCESS in 8m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/60e3f716f39d4076a4bbbaa3d33b6f8b : SUCCESS in 12m 46s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ba9e1cd688be4650aed3ec3e93520604 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1a9ea35916774679aaaf73e414c49cd1 : SUCCESS in 6m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/71790e7b9bb74ef9b5711de1fc8707e8 : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c3257b2d53174033a9018e3ebb0e7e11 : SUCCESS in 45m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b4da78876db0404f9152cc24dbe30274 : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3435da3c39714bc5843082238fe288d1 : SUCCESS in 1h 20m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b617f4e3183345fab954b4040aa2e11c : FAILURE in 2h 14m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/53bd4e0eb5be4129897b178e7d1dae7f : SUCCESS in 1h 28m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/244fb6ef70f7420c8f52bdb1d41870f9 : SUCCESS in 1h 09m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c3c0321d30c14931b50f716bbe3475a2 : SUCCESS in 1h 14m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8be98e8c563d47a6b6583f536b091d03 : SUCCESS in 1h 32m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/34ec5a147fd74ae192c035962e87718b : SUCCESS in 2h 28m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6d85f3e9cde45ba85ecfb27a21f4caa : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":1},{"id":"83d7416706407f0a196595be3899dc72f722e4b2","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-04-06 11:50:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/53/836753/1/upstream-solidfire-check/cinder-solidfire-iSCSI/0c54390/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"daf910e7b4efce76b82c17fe61c7220c619f798c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-06 11:54:09.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/5c2fd1453cc34cf68fa3104a754ed2a0 : POST_FAILURE in 2h 36m 33s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/b4f5d0394018465abff10768e6524b10 : POST_FAILURE in 1h 19m 47s","accounts_in_message":[],"_revision_number":1},{"id":"160e0a034471ddfa5e4ba360a8b7c0361601aa9c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-04-06 11:55:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/53/836753/1/upstream-check/cinder-cDOT-iSCSI/0738cf8/ : FAILURE in 3m 55s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/53/836753/1/upstream-check/cinder-cDOT-NFS/feb2b2c/ : FAILURE in 4m 20s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/53/836753/1/upstream-check/cinder-cDOT-FCP/6d7f6ea/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"3720adca736539f3b468749d949375f5d1556127","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-06 12:06:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/53/836753/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/06a9440/ : FAILURE in 2h 42m 29s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/53/836753/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/123cf42/ : FAILURE in 2h 51m 37s","accounts_in_message":[],"_revision_number":1},{"id":"acca8d482d114de8b270aa82bbc055b4534084db","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-06 12:27:36.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-53-836753-1/ns5_nfs : SUCCESS in 3h 14m 32s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-53-836753-1/ns5_iscsi : SUCCESS in 3h 14m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"718ea2c5db8fdf12a40d67784fbe535f5f07a87d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-04-06 14:56:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/53/836753/1/ : FAILURE in 1h 42m 37s","accounts_in_message":[],"_revision_number":1},{"id":"95d620952a6113189fe2daf1991a402c79fc172c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-06 19:06:54.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-53-836753-1/ns5_nfs : ABORTED in 3m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-53-836753-1/ns5_iscsi : ABORTED in 3m 49s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"2e3d4ab3d409e4a334190d940d23c401e7e7d600","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-07 06:27:51.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-53-836753-1/ns5_nfs : SUCCESS in 3h 18m 4s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-53-836753-1/ns5_iscsi : SUCCESS in 3h 24m 20s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"1db5e2b80c9da3d545307a16840d3d2554f36480","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-04-08 05:14:27.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ed7db2aa1770ce7f4905b41779a2cabe14c2d532","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-04-20 13:49:48.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f539f1791d6efbe73070559d62e213e6c5f089a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-04-20 14:54:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b66bf9167a8a23a7be6ca06ddbb833ac6e71717","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 14:55:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ba2f048a883da95033b0f117532fd4840a9fd6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 17:18:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f23b3175590a4d95b91e5bc26eb1989f : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aeac55ccf906465b84cd4b540f956cae : SUCCESS in 9m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eeaca8e951964f608e2bf4c1952da7a0 : SUCCESS in 10m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0828f124d0b54260980b451e51f10546 : SUCCESS in 17m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/46aedee23d884150a8f22857ea905264 : SUCCESS in 55m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ebd31eed8b840e3af878abe9f15887e : SUCCESS in 1h 23m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ffcbcfc94da4ebbbb74e867c883235d : SUCCESS in 56m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6cc545b75244a75833ea2f54da75968 : SUCCESS in 9m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7d685e55328146ba9cee63c540e26542 : SUCCESS in 1h 11m 29s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/863bb9caac5f4cdea7ea3ede08f2e30a : SUCCESS in 1h 36m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0740f703cbe34a83ab14337a88657bf2 : SUCCESS in 1h 03m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b09ca1fc4fc593ab383b9dcac04f635b686cda2b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 17:18:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"42923f049be453a3d257c872eb074c46110d869e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 17:22:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1d003e63627409d894e4a40c86dd947 : SUCCESS in 1m 34s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b1c58aaad804bca94f3c00875ca5db6 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7bb194d092a9f48c09b539ef6153294273147e5","revisions":{"d7bb194d092a9f48c09b539ef6153294273147e5":{"kind":"REWORK","_number":1,"created":"2022-04-06 09:12:32.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/53/836753/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/53/836753/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/836753/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/836753/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/53/836753/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/53/836753/1"}}},"commit":{"parents":[{"commit":"a6ffb102af7625c53da40358268955b177516aa5","subject":"Merge \"Reject bad img formats for uploaded encrypted vols\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a6ffb102af7625c53da40358268955b177516aa5"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-03-26 17:05:11.000000000","tz":-240},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2022-04-06 08:01:42.000000000","tz":120},"subject":"Fix: Schema validation for attachment create API","message":"Fix: Schema validation for attachment create API\n\nCurrently the schema validation for attachment create assumes that\ninstance UUID will always be present in the request but that is\nnot the case when glance calls cinder for attachment.\nAlso there isn\u0027t any schema validation for MV3.54 which accepts\nattachment mode in the request hence failing requests passing mode.\nThis patch removes the instance_uuid from required parameters and\nadds schema validation for MV3.54.\n\nAlso includes a squash to add a releasenote from\nChange-Id: I4a6e93ea98cfd4988d38bedca6e5538391c1f74d\n\nChange-Id: I5108fd51effa4d72581654ed450d191a13e0e964\n(cherry picked from commit 560318c82e9951464361560e57ce0b67011cce74)\n(cherry picked from commit 58250aae06988858a27576f6d471d7f37b6dcecb)\nConflicts:\n    cinder/api/v3/attachments.py\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7bb194d092a9f48c09b539ef6153294273147e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7bb194d092a9f48c09b539ef6153294273147e5"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
