)]}'
{"id":"openstack%2Fcinder~836944","triplet_id":"openstack%2Fcinder~master~Ic97a2501b4c895b2b655284bcc320036c2d4e217","project":"openstack/cinder","branch":"master","topic":"bug/1968159","attention_set":{},"removed_from_attention_set":{"32074":{"account":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"last_update":"2022-04-20 17:18:46.000000000","reason":"Change was submitted"},"34541":{"account":{"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},"last_update":"2022-04-19 14:27:00.000000000","reason":"Manish replied on the change"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-04-19 12:25:03.000000000","reason":"Gorka Eguileor replied on the change"}},"hashtags":[],"change_id":"Ic97a2501b4c895b2b655284bcc320036c2d4e217","subject":"[SVf]:Fix retype failure for replication volume-type","status":"MERGED","created":"2022-04-07 11:38:22.000000000","updated":"2022-05-03 09:09:00.000000000","submitted":"2022-04-20 17:18:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836944-bug/1968159","meta_rev_id":"e492fcbaacea512b65447436f48d56cd9a88c0d0","_number":836944,"virtual_id_number":836944,"owner":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-20 17:18:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"recommended":{"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},"all":[{"value":1,"date":"2022-04-20 07:35:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-19 14:27:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},{"value":2,"date":"2022-04-20 15:13:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2022-04-19 12:25:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},{"value":1,"date":"2022-04-20 15:13:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}],"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},{"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-07 12:02:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-07 12:36:29.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-04-07 12:43:11.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-07 13:21:19.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-07 13:43:49.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-04-07 14:01:52.000000000","updated_by":{"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},"reviewer":{"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},"state":"REVIEWER"},{"updated":"2022-04-07 14:09:54.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-04-07 14:53:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-07 15:11:09.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-04-07 16:29:45.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-07 18:21:38.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-07 21:06:49.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-07 22:01:18.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-04-08 07:53:52.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-09 15:25:10.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-04-10 06:44:54.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-11 11:21:12.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-04-12 08:51:34.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2022-04-12 10:53:51.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-04-12 13:13:54.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-04-12 19:08:46.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-12 21:20:34.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-04-12 21:49:36.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-04-12 22:19:08.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-04-12 22:27:05.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-04-18 18:49:16.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2022-04-19 08:05:41.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-04-19 09:10:40.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-04-19 09:43:04.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-19 14:47:16.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-20 07:35:59.000000000","updated_by":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"reviewer":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"state":"REVIEWER"},{"updated":"2022-04-20 15:13:03.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"dcf11145108f32a5bc5f05c51718ffa0df21072d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"date":"2022-04-07 11:38:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54c4601329ec8d0bc589ef53406e8ea7ab8b03a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 12:02:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/17968f7fab3c42b19924c56e61cdd5c0 : FAILURE in 20m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29cf7ebe774548b0aac54a865b5f55a8 : FAILURE in 22m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90060efdd711f7f9de9a6938cb93b3ae3747d05b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"date":"2022-04-07 12:11:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2dcff81c807cd97286451dce232aaf7aefad04e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 12:35:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/36acda40c7cb497295258de81a593ce8 : FAILURE in 19m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8cc9188783ab4f5f8d90837de258a8e3 : FAILURE in 21m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cef7311be06098a6acc3ecc6c3c0d7573522938c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-07 12:36:29.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/44/836944/2/check/DellEMC_Unity_FC/74d9d1f/DellEMC_Unity_FC/524 : FAILURE in 19m 39s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/44/836944/2/check/DellEMC_Unity_iSCSI/3417b70/DellEMC_Unity_iSCSI/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"99225f2b0802e3dbbd043ed9bc97ad5003a7b098","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-07 12:43:11.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15751/","accounts_in_message":[],"_revision_number":1},{"id":"9d8080231851b7c98cd557a981a6f0ce5accfd79","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-07 12:44:55.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15751/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"085d0d56d3ae363e868dea60d4e3416d08bcab58","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-07 12:46:22.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15752/","accounts_in_message":[],"_revision_number":2},{"id":"868dbc080c697c7e40f6d31b7b6f077f3acb749a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-07 12:47:54.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15752/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"c5ed00f7ff6e723c2d46ce48385d4255149a3890","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-07 13:21:19.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/836944 : FAILURE in 57m 31s","accounts_in_message":[],"_revision_number":2},{"id":"68e54b8bb514f9a3ddc2443930b18622e3817675","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-07 13:25:42.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/836944/836944/2/check-not-vote/ext-cinder-zuul/42a1589 : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":2},{"id":"5594550c7ca8955c4dde3f5fed5190e2a8fb173d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-07 13:43:49.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/836944/2/iscsi/ : FAILURE in 1h 31m 27s\n- cinder-smb http://cloudbase-ci.com/cinder/836944/2/smb/ : SUCCESS in 55m 47s","accounts_in_message":[],"_revision_number":2},{"id":"d10c5762d0a8dda7f91d32ac3ac657cc2c394066","author":{"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},"date":"2022-04-07 14:01:52.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d098cf73f319ebc8ca2c5208d79a45defdb4d64f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-07 14:09:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/44/836944/2/check/iscsi-driver-alletra-6k/a9d6457/a9d6457.html : FAILURE in 1h 02m 08s","accounts_in_message":[],"_revision_number":2},{"id":"9dbfb17d90426b2f8d1f55c972d42d1ff9894c75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 14:53:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86aea14f36ae4f0c840201e5e5960099 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eaddcd3c3c37426f874bbd79dbffd671 : FAILURE in 10m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9dfce20ec443436482eaa1a457e64897 : FAILURE in 12m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/767c5ea42b8645dd8d6caeaf280b56f8 : RETRY_LIMIT in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/991a1cbd383442aa928d2d8e878cd8f1 : SUCCESS in 13m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/20d956dd60dd40ea95ae20cbfadacfef : SUCCESS in 59m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/38471bfe5950470b8b20bdb785f26235 : SUCCESS in 1h 29m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80ca7d4b3fce4e20b64e5d6f399da626 : SUCCESS in 53m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c838d5ab0b64fcab90d280a1cf9f508 : SUCCESS in 8m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9f213d29aa1c4555962f814b859fada0 : FAILURE in 12m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/49efba7d05204046849b976897e451ae : SUCCESS in 5m 39s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e6816c38b174e5dafefe792aab04f4d : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cb065b8f7a124ae78d0a3b3c31b9ad87 : SUCCESS in 6m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/af1e0482ea96486fa125d1bbe7733a5e : SUCCESS in 7m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/766805b2fa254f07a447840e582c5c3c : SUCCESS in 47m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d9bdedc345684b5492c4c68a32ce83bb : FAILURE in 4m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/32620a755939465f8010b4becb7811c9 : SUCCESS in 1h 18m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dd3db01a59ec4ddb8e556c7049098249 : SUCCESS in 1h 38m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/41975446447546928cc69017fd174417 : SUCCESS in 1h 45m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c8e2dd5f89fc45d8b59b436fcf440244 : SUCCESS in 1h 17m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3135e9ff546d4faeae953ccbe4a12ffc : SUCCESS in 1h 11m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6ba0bd04c41a4af180b7b1ac876a786b : SUCCESS in 48m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dce125de66c44f09a22869bf462892e5 : SUCCESS in 1h 26m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8f795b1312d8452da8fc4290d7e0464c : SUCCESS in 2h 34m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2512e141f234b26900ada54ab7a58cb : SUCCESS in 48m 27s","accounts_in_message":[],"_revision_number":2},{"id":"da19b7f4f2919cb43522ebb13f4cc4bf548ae023","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-07 15:11:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/44/836944/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/da83a59/ : SUCCESS in 2h 26m 05s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/44/836944/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f1a0098/ : SUCCESS in 1h 28m 34s","accounts_in_message":[],"_revision_number":2},{"id":"2dad4dcba805afa6901660b9f80d88057d0acb83","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-07 15:16:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/836944 : SUCCESS in 1h 22m 11s","accounts_in_message":[],"_revision_number":2},{"id":"8a1dad72d8b46baf905fc572d7564ebe80570475","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-07 16:29:45.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-44-836944-1/ns5_nfs : SUCCESS in 3h 32m 8s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-44-836944-1/ns5_iscsi : SUCCESS in 3h 32m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"e871cc4264bf47bfa4541bb90184f86e6060fa3e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-07 16:53:44.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-44-836944-2/ns5_nfs : SUCCESS in 3h 26m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-44-836944-2/ns5_iscsi : SUCCESS in 3h 25m 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"297cf3b191a7c6efc50d5363e6b7aea7554f970e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-04-07 18:21:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0629ad1cb4ec4072b277122a9547ea8b : SUCCESS in 1h 29m 51s","accounts_in_message":[],"_revision_number":2},{"id":"230b30db443fabc4adc1d652e79d9e9a677363c5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-07 21:06:49.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/bdb161c2c1384876ab29101d858b8c44 : POST_FAILURE in 1h 59m 14s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/f09caab3b7d34b44bcebdea1e730e269 : POST_FAILURE in 2h 37m 39s","accounts_in_message":[],"_revision_number":2},{"id":"4e70dde139b27a6bfb56b60f45a28a7fe2402c99","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-07 22:01:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/44/836944/2/check/dsvm-tempest-cisco-zonemanager-job/113b50d/ : SUCCESS in 1h 12m 00s","accounts_in_message":[],"_revision_number":2},{"id":"acc81794c5eb429f26d7e3f8a247ed7f57473e03","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-04-08 07:53:52.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/44/836944/2/ : SUCCESS in 2h 15m 35s","accounts_in_message":[],"_revision_number":2},{"id":"b8885fd989fa05462383984d2a196840e878b85b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-08 12:42:39.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/44/836944/2/check/iscsi-driver-nimble/9b011fc/9b011fc.html : FAILURE in 1h 01m 01s","accounts_in_message":[],"_revision_number":2},{"id":"eaa1069186bd1baeae6d0454571b3b330dac90f9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-08 13:44:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/44/836944/2/check/iscsi-driver-nimble/4b1ac7c/4b1ac7c.html : FAILURE in 1h 00m 11s","accounts_in_message":[],"_revision_number":2},{"id":"fbb1fdaef9c891ec69aabce5de01b39d8ea16c1a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-04-09 15:25:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/44/836944/2/20220409220403/NEC-V-FC : SUCCESS in 1h 57m 44s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/44/836944/2/20220409220403/NEC-V-ISCSI : SUCCESS in 2h 21m 05s","accounts_in_message":[],"_revision_number":2},{"id":"cd8eec2e0b0f6afe49686c8064e1702db975352f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-04-10 06:44:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/44/836944/2/20220410122418/NEC-FC : SUCCESS in 3h 20m 29s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/44/836944/2/20220410122418/NEC-ISCSI : SUCCESS in 3h 20m 33s","accounts_in_message":[],"_revision_number":2},{"id":"8f10dc025a8a69de8bac704070803f312b79813c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"date":"2022-04-11 11:21:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4a2f8c61a6162a6de63afb67453e551f545ecb83","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-11 11:21:12.000000000","message":"Patch Set 3:\n\nBuild Started","accounts_in_message":[],"_revision_number":3},{"id":"8c825afc451ec014029e8a2db0d3375fe3787cb3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-11 11:21:14.000000000","message":"Patch Set 3:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15772/","accounts_in_message":[],"_revision_number":3},{"id":"91c55abc2796f914f718d2071fa1ae8157a92bb0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-11 11:22:59.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/44/836944/3/check/DellEMC_Unity_FC/21e72ab/DellEMC_Unity_FC/549 : FAILURE in 1m 40s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/44/836944/3/check/DellEMC_Unity_iSCSI/b02cd3f/DellEMC_Unity_iSCSI/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"6a1b89ec7abbba78f8a524a0f63111b9fed69b29","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-11 11:23:05.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15772/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"c447d0af67cf335d0aa2479aa8f2d842156ccbd6","author":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"date":"2022-04-11 11:35:31.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b44d881ff9cf8af09c397f0259c0fffc5e641299","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-11 11:47:31.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1649676071241 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"959a008c3d455484d16003e9916a9945e2e6127c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-11 12:19:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/836944 : FAILURE in 56m 54s","accounts_in_message":[],"_revision_number":3},{"id":"0d9e960c2765ff7f911e75766574f7ed004fd613","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-11 12:25:10.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/836944/3/iscsi/ : SUCCESS in 1h 03m 28s\n- cinder-smb http://cloudbase-ci.com/cinder/836944/3/smb/ : SUCCESS in 1h 00m 15s","accounts_in_message":[],"_revision_number":3},{"id":"43edec0e800d2c8aff9c06cd630f588088c90415","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-11 12:26:16.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/44/836944/3/check/iscsi-driver-alletra-6k/bec30b7/bec30b7.html : FAILURE in 1h 03m 49s","accounts_in_message":[],"_revision_number":3},{"id":"285e600cdff8a1da8270be24a42734c477bead5a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-11 12:28:17.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/44/836944/3/check/iscsi-driver-nimble/c2b2d9d/c2b2d9d.html : FAILURE in 1h 05m 59s","accounts_in_message":[],"_revision_number":3},{"id":"bc623a495303b1661553336535d57d658f795cbd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-11 12:43:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/836944 : SUCCESS in 1h 20m 00s","accounts_in_message":[],"_revision_number":3},{"id":"78e5d9c18e61138eb201269a78b25e1bf258fba0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 12:45:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eef3e3d1b17547419d97685bdac56617 : SUCCESS in 23m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/38cd4dd9ccd74d59828fc4017373786f : SUCCESS in 25m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bb06faf42408080d55c501d65739771e76c323d1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-11 12:47:39.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/44/836944/3/check/dsvm-tempest-cisco-zonemanager-job/8e34580/ : SUCCESS in 1h 13m 31s","accounts_in_message":[],"_revision_number":3},{"id":"0e9f63c6037f2f5399e4e38b453f8801b90427a8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-04-11 12:53:01.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2eedb7a969cd4b1e91370c753a76fc5a : SUCCESS in 1h 30m 47s","accounts_in_message":[],"_revision_number":3},{"id":"277562b7175df3e28db76d300b7a45ee1b962e46","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-04-11 13:28:16.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/44/836944/3/20220411202105/NEC-V-FC : SUCCESS in 1h 48m 02s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/44/836944/3/20220411202105/NEC-V-ISCSI : SUCCESS in 2h 07m 08s","accounts_in_message":[],"_revision_number":3},{"id":"052af721261d3563713f80512d14225d98d0dbc6","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-04-11 13:34:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/44/836944/3/ : SUCCESS in 2h 03m 11s","accounts_in_message":[],"_revision_number":3},{"id":"ae356f77ee801764985f93a4711d99359bd5602c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-11 13:46:59.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/44/836944/3/check/iscsi-driver-nimble/3998c8a/3998c8a.html : FAILURE in 1h 16m 56s","accounts_in_message":[],"_revision_number":3},{"id":"cc552051a4e8593224f1b77e61d57f2a52917b9d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-11 13:55:55.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/44/836944/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5c2a286/ : SUCCESS in 2h 32m 34s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/44/836944/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/aa4672d/ : SUCCESS in 1h 10m 14s","accounts_in_message":[],"_revision_number":3},{"id":"913773a49b0be6cb0fad1dfdbbf4a51166496709","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 13:58:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/849b77f4283f4b369135f5c9ab385a86 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24f66f77056f495d83fbf85f08f6d751 : SUCCESS in 11m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da74b3723e1241c7a47a92ea8c903c88 : FAILURE in 12m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6c6948c948b40ab857d49088dd0ca76 : RETRY_LIMIT in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d00ce4da63f34a469a65244ce63252d1 : SUCCESS in 13m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/5c7f3196b5d24fd3b54c4c4a7ab0c02f : SUCCESS in 1h 09m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c793a77156bb4c12a16b7df6bef720f1 : SUCCESS in 1h 18m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff0e6cd54a1141829f14acfa7f0c496a : SUCCESS in 59m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe2f12f9c93347a8ba763f849302f4a5 : SUCCESS in 12m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cad588792e6b43b3ba1589a725e6a139 : SUCCESS in 19m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/671e9cc78f1e431d8ea4fe9e490dbb06 : SUCCESS in 9m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f72b8825b33840d48561417258f1fa0e : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/499ff96f9caf4936b34820fea9cce1a6 : SUCCESS in 8m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f1673b35c1f142e0a4ffb94288bde608 : SUCCESS in 10m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e716673d6b414dd386e6d84931889496 : SUCCESS in 44m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/caf2f980d9aa4068b759869037830474 : FAILURE in 7m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f66b0f2803024612b34de2626151fe4a : SUCCESS in 1h 20m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/249ced1503ac4b5fa77d904f846725cc : SUCCESS in 2h 02m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5b325128e8b04809ab32e5247fdbdef3 : SUCCESS in 1h 55m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cd921aea1a8b408e84907ed513386fd0 : SUCCESS in 1h 11m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/76a470937c284c69a89e11f393f8b31f : SUCCESS in 1h 30m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b5ae2911be8c4c9ca7ad851e44cb1f7b : POST_FAILURE in 38m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/628566f5be9c485fa618004fca4af393 : SUCCESS in 1h 23m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/40fcbd9a9b2348f69c8fb09f0bbb3d1e : SUCCESS in 2h 16m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52fc925b46d248d5b56900e3128f8a07 : SUCCESS in 49m 26s","accounts_in_message":[],"_revision_number":3},{"id":"71f43581ef3f3b18906d1a9fa76bb725c2d5f344","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-11 14:17:39.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/61464af98de34c79b671e0d041fae5ec : POST_FAILURE in 2h 00m 49s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/b6ba41ab00a44cc2a90eab816de72cd3 : POST_FAILURE in 2h 53m 27s","accounts_in_message":[],"_revision_number":3},{"id":"7df0740395cba6e4e0ffc8561c3b427f3a9cb670","author":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"date":"2022-04-12 07:25:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6901d6a6cbcb99d8f222230d4b58235483c46ec5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 07:48:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/880a2b232b3f48619c5af3e466b90011 : SUCCESS in 19m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ecd7e94c882e490f82c2d4744bf10ed2 : SUCCESS in 21m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"941c5c7f2bca84358f8856a33b9bf50d21c01fde","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-12 08:08:55.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-cinder http://207.189.188.190/logs/836944/836944/3/check-not-vote/ext-cinder-zuul/a7b014a : SUCCESS in 42m 36s","accounts_in_message":[],"_revision_number":3},{"id":"f5237872eb7a2245ab6dda8edf1554e92b633186","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-12 08:32:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/836944/3/iscsi/ : SUCCESS in 1h 06m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/836944/3/smb/ : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":3},{"id":"0f8085e03b5562d91d468bf3fc50ce0d7535f42c","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-04-12 08:51:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/b06862e/index.html : SUCCESS in 1h 18m 46s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/c064197/index.html : SUCCESS in 1h 25m 41s","accounts_in_message":[],"_revision_number":3},{"id":"a84b17bd125e5da75517652dd2aa366e6299aead","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-04-12 09:49:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/44/836944/3/20220412121517/NEC-FC : SUCCESS in 3h 18m 58s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/44/836944/3/20220412121517/NEC-ISCSI : SUCCESS in 6h 34m 23s","accounts_in_message":[],"_revision_number":3},{"id":"bf364375b8d7667f927988b2e8769ee129713c38","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-12 09:53:40.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/d1440e33a9244322a2a9128dc106e444 : POST_FAILURE in 1h 36m 33s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/26497c7a8da3467cb804c2bae8ff53b5 : POST_FAILURE in 2h 26m 29s","accounts_in_message":[],"_revision_number":3},{"id":"843f0f5ba830cc910017224d76a857cc1619eb91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:21:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21f8795e4fbc4371839924f093e902e6 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f21ff1263ec64ccfb7e9399848b25231 : SUCCESS in 10m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef1153b2d46f4e938cba68b053b0ec6f : SUCCESS in 11m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac188f8723ce4292b9c297850c1bbc10 : RETRY_LIMIT in 3m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/105fa644e6d249fbabfeac31c75aff59 : SUCCESS in 12m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/2f5c0d5bca40450f9ae81a5ebfece3aa : SUCCESS in 1h 08m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eedeb82e46c04f54bdadb5763489972b : SUCCESS in 1h 10m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/20627ddc04cc4a8196e16694505463c3 : SUCCESS in 55m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/766b7a8c477e4cf0950d65e9764ae22f : SUCCESS in 8m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/98249baf41ae4239a79581c211057f3e : SUCCESS in 14m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a91391357ee44fc0aaacbffa826dacce : SUCCESS in 6m 08s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c2c787551715490d98f383d189f85bad : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1d167f5699014137bcbc87d70269e671 : SUCCESS in 7m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6a90cb96a08d45b0bd8aeab8a4e88582 : SUCCESS in 7m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0d1906ed33ac40aa81ed8d1c0157f23e : SUCCESS in 41m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/34bfa83171a44d6fa3fb57b9fe6e46c0 : FAILURE in 4m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/75fe65750cb3414cbc0756f205e5bbc6 : SUCCESS in 1h 28m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e9fa09bc4d76419d893fc412a12357f3 : SUCCESS in 1h 28m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ea574dd8bc274bbfa84ae0f05b207680 : SUCCESS in 2h 00m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ee3d811b5f8448169f8b17e72fa40e94 : SUCCESS in 1h 12m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/03125bd9b39445459cab6c6425d45a46 : SUCCESS in 1h 33m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4d4cd711633641df80596ae3e63c81f5 : SUCCESS in 44m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/25983f341a3946afb092d6a333be0bcb : SUCCESS in 1h 18m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c31a858858ba43e0959324d8a66cef76 : SUCCESS in 2h 47m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26c2715824ef434180487c55cb3a7251 : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":3},{"id":"6815ab53c9cd6a76172f401320950d02005ea340","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-04-12 10:53:51.000000000","message":"Patch Set 3:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/12_04_2022_10_21_38\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"ae745a281c8eae10936b772ff109d3b557ac6e12","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-04-12 13:13:54.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/44/836944/3/check/DellEMC_SC_FC/f17629e/DellEMC_SC_FC/450 : SUCCESS in 1h 10m 24s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/44/836944/3/check/DellEMC_SC_iSCSI/084dabb/DellEMC_SC_iSCSI/471 : SUCCESS in 1h 52m 19s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"da28bbb7188a623c136d6001256607fac00dc694","author":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"date":"2022-04-12 16:54:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b8a534349b28031e42d8291542a72409f6851357","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-04-12 18:22:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/072e10a/index.html : SUCCESS in 1h 22m 24s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/42870ae/index.html : SUCCESS in 1h 27m 11s","accounts_in_message":[],"_revision_number":3},{"id":"18dc3dfb033af81a4a5d85d4d535cfaadabefac3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-12 18:48:01.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/44/836944/3/check/dsvm-tempest-cisco-zonemanager-job/91ff536/ : SUCCESS in 1h 13m 52s","accounts_in_message":[],"_revision_number":3},{"id":"b27787964b5942814b7ac28e1b2d659ff599b513","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-04-12 19:08:46.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/44/836944/3/check/DellEMC_PowerFlex/0fe2b24/DellEMC_PowerFlex/578 : SUCCESS in 1h 33m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"8c741973c2425c6f2b58b6e51c4621e6301324b5","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-04-12 21:20:34.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"a11be6dbc3127805130c0c4f5b081b4c2e1ee93d","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-04-12 21:49:36.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"f461a4550c56f7eba13343e816533b612b84d6b3","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-04-12 22:19:08.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"7d62f878cd264a0796dc4c1662ff409eecf7b13b","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-04-12 22:27:05.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"9515ab3ed5d863cc7e7fc9d61f91d0297a051477","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-04-18 18:49:16.000000000","message":"Patch Set 3: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bdb0636e4642428641c96ca6aa0e25df22cda82b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"date":"2022-04-19 05:20:51.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"708a061c96d657d3c03b36a8bd433b81b877eb08","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-19 05:21:01.000000000","message":"Patch Set 4:\n\nBuild Started","accounts_in_message":[],"_revision_number":4},{"id":"ab735bdd61cfd083cfdcd8b1c32da35d06095fb3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-19 05:21:02.000000000","message":"Patch Set 4:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15839/","accounts_in_message":[],"_revision_number":4},{"id":"ea9312e0d0acb09e22b79ab2f841c06819519a27","author":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"date":"2022-04-19 05:22:08.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2ebe8575978f6b8467f1d5e4dd14e715081ce57b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-19 05:22:56.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15839/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"fd816f173fb4130b3949a512c7d75047e5ca0416","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-19 05:24:23.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1650345660878 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"f6327e82f5ae42ca80f82b5fff419730d852241f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 05:46:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0fbbea3bb1d4aedad398de5a82713fb : SUCCESS in 21m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9db79357638c4b5191b5e387f54b3e40 : SUCCESS in 23m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b055ea28302e3e7db1bfd650f0535f36b1f723cc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-19 05:56:01.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/836944/4/iscsi/ : FAILURE in 34m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/836944/4/smb/ : FAILURE in 31m 18s","accounts_in_message":[],"_revision_number":4},{"id":"5cf192b7cf40df814c7b914fd39d3f16439ebefe","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-19 06:43:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/44/836944/4/check/dsvm-tempest-cisco-zonemanager-job/134363e/ : SUCCESS in 1h 15m 40s","accounts_in_message":[],"_revision_number":4},{"id":"72b34e05b5aea1e95ba0b984c4324cf17eaffba8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-19 06:53:42.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/44/836944/4/check/DellEMC_Unity_FC/e3a04c8/DellEMC_Unity_FC/627 : SUCCESS in 1h 09m 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/44/836944/4/check/DellEMC_Unity_iSCSI/6a8016e/DellEMC_Unity_iSCSI/524 : SUCCESS in 1h 32m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"c653b4ffa6b54286706a5ac3e338caa3560933dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 07:47:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a0d873b674447bcb2ad253f7f4174bc : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ef7984687ec448bbc41edd2acb20270 : SUCCESS in 10m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69c554e341df416c844c7c9f66974ff1 : SUCCESS in 10m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78dd87788f7943188a2e1921762a1908 : RETRY_LIMIT in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/186ebf9835354cffa582025bb85c54d0 : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/d253a34d8a2d414b983a1227ab7c8d43 : SUCCESS in 45m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/03052c006a314096a0239d73a30659c8 : SUCCESS in 58m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b0de327332a40e28f4922f1f767229f : SUCCESS in 38m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a95b4f3ddd57443da5867bb7f69e141a : SUCCESS in 8m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/397cda62890c453686b9b4fcb6b298d0 : SUCCESS in 12m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/182194ce071f413cb87ab593de3e7667 : SUCCESS in 7m 12s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8675aeec5a9645968652d4b3105917d5 : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7e5247ca897c48ad9c263fad43abbdb3 : SUCCESS in 6m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0ba4ab1e999448a09c07d079c91ee528 : SUCCESS in 6m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8e64fcbc4b6040c08832547f8851aac6 : SUCCESS in 46m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/762da55cbe9942e9917a451e1d8b164d : FAILURE in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d5042327a5084fabac0b47248c4d7b06 : SUCCESS in 1h 15m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0eff5d11475544ef96505c22901f599d : SUCCESS in 1h 11m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b28f488f4c1a46438b046631bc362840 : SUCCESS in 1h 29m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/55acab40b24448c9bda2a793a5d32595 : SUCCESS in 1h 21m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/58323af611ba4f1d9460dd0014824f58 : SUCCESS in 1h 11m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/636a64aba3764de48d04ddab78a032c9 : SUCCESS in 32m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dcb7336c462744cba40f1502cc5828b7 : SUCCESS in 1h 04m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5e2c2c330b1a4df2b73d33c27237ffb8 : SUCCESS in 2h 24m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf3a979400c440cb96aafb2fc845ec62 : SUCCESS in 1h 09m 32s","accounts_in_message":[],"_revision_number":4},{"id":"258121b6f6d8cdf19e9b33845aa0e2c78477732f","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-04-19 08:05:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-44-836944-4/1650354439 : SUCCESS in 18m 20s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"44e65a1c7779ad26807ce3e6f50660d9304af1eb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-19 08:25:51.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/44/836944/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5664bc6/ : TIMED_OUT in 3h 02m 52s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/44/836944/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/aa74950/ : TIMED_OUT in 3h 02m 50s","accounts_in_message":[],"_revision_number":4},{"id":"8b7c34a7c8935393da5ebb1eb29c1a77ff9b23ca","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-04-19 08:28:39.000000000","message":"Patch Set 4:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/19_04_2022_07_47_26\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"8f23575440cb698a03ebec6fbb5c9b5fca9e4f84","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-19 08:44:06.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-44-836944-4/ns5_nfs : SUCCESS in 3h 22m 45s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-44-836944-4/ns5_iscsi : SUCCESS in 3h 20m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"0ab17da2b527166c1690d6d64ae3acd758e5f7e4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-04-19 08:52:19.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/44/836944/4/check/DellEMC_PowerFlex/fb1bc7d/DellEMC_PowerFlex/621 : SUCCESS in 1h 04m 41s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"b3e8ebd010f3de3b36462e4b89a9bd7ebfcd3d9d","author":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"date":"2022-04-19 08:58:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"327424c086ce44d3b8075a24349273e482bfa551","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-04-19 09:10:40.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/44/836944/4/check/DellEMC_PowerStore_iSCSI/8a86d0a/DellEMC_PowerStore_iSCSI/417 : SUCCESS in 1h 23m 00s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/44/836944/4/check/DellEMC_PowerStore_FC/26f38a5/DellEMC_PowerStore_FC/432 : SUCCESS in 1h 14m 54s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"e6a6dceb1c80bf2b502845fcc070a3751233b266","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-04-19 09:34:33.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/44/836944/4/check/DellEMC_SC_FC/523bb90/DellEMC_SC_FC/494 : SUCCESS in 58m 50s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/44/836944/4/check/DellEMC_SC_iSCSI/708f7b9/DellEMC_SC_iSCSI/515 : SUCCESS in 1h 46m 57s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"2b10014e1911686db0ae741daf05f91e03f8ef2b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-04-19 09:43:04.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/44/836944/4/check/DellEMC_VNX_iSCSI/639c285/DellEMC_VNX_iSCSI/284 : SUCCESS in 1h 55m 27s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/44/836944/4/check/DellEMC_VNX_FC/e26024e/DellEMC_VNX_FC/300 : FAILURE in 18m 40s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"a57ebdae553da6d12f9c7a7192c8a706b2ce68fd","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-04-19 09:58:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/836944-4/HPMSA-iSCSI/ : SUCCESS in 53:16.84\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/836944-4/HPMSA-FC/ : SUCCESS in 45:46.29","accounts_in_message":[],"_revision_number":4},{"id":"005b26a6f25c391c3d92726b247e9a6047de0a5c","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-04-19 10:22:48.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/836944-4/Lenovo-iSCSI/ : SUCCESS in 1:04:57\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/836944-4/Lenovo-FC/ : SUCCESS in 58:47.85","accounts_in_message":[],"_revision_number":4},{"id":"aff7995fa80904d05c99bbfd342f744331c3bcc4","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-04-19 10:35:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/836944-4/Seagate-iSCSI/ : SUCCESS in 1:10:57\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/836944-4/Seagate-FC/ : SUCCESS in 1:05:31","accounts_in_message":[],"_revision_number":4},{"id":"1e7bb204201f8aa3af0aece7ab007f3e3c103f89","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-04-19 10:38:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/836944-4/PVME-iSCSI/ : FAILURE in 1:14:17\n- PVME-FC https://os-logs.seagate.com/dell-ci/836944-4/PVME-FC/ : SUCCESS in 1:03:26","accounts_in_message":[],"_revision_number":4},{"id":"b55bfc8f0cb3c41edd874ed347f99709d62484f4","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-04-19 11:35:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/811b67c/index.html : SUCCESS in 1h 11m 21s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/214f443/index.html : SUCCESS in 1h 15m 45s","accounts_in_message":[],"_revision_number":4},{"id":"405ef911dcedac2d8ccc053b056501ed01e4c47c","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-04-19 12:25:03.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"844fa1a88a60aa173d67be7a8ebd8eab9a358c9d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-19 13:32:55.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/44/836944/4/check/DellEMC_Unity_FC/8f6880e/DellEMC_Unity_FC/631 : SUCCESS in 1h 08m 28s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/44/836944/4/check/DellEMC_Unity_iSCSI/169a76a/DellEMC_Unity_iSCSI/528 : SUCCESS in 1h 43m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"bf011bc23f3ccc41f89a52ba60e2f50497bac976","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-19 13:52:15.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/44/836944/4/check/tempest-dsvm-full-focal-py3/a9661b0/ : SUCCESS in 2h 45m 37s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/44/836944/4/check/devstack-plugin-ceph-dsvm-focal-py3/22add3f/ : SUCCESS in 3h 08m 15s","accounts_in_message":[],"_revision_number":4},{"id":"eef3fb4cd45c2b5f52422715d9bb3a674b21ba57","author":{"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},"date":"2022-04-19 14:27:00.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9e38bf08848b4a091a5c28d333b3c72ec80b97da","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-04-19 14:47:16.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/44/836944/4/check/DellEMC_PowerMAX_FC/69ae750/DellEMC_PowerMAX_FC/431 : SUCCESS in 3h 19m 46s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/44/836944/4/check/DellEMC_PowerMAX_iSCSI/8eb64d3/DellEMC_PowerMAX_iSCSI/438 : FAILURE in 35s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"44049d5e176fa0d5633e7392b7988eb44d003547","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2022-04-20 07:35:59.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0694b8e2f82dc5b7ca840246d99d07edd9ab20b9","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-04-20 15:13:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"81267a49fb2e7989f8f55db8d0c8132376d35e18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 15:14:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"388186ae71ccbc17956d63d5ef8d2a0c76edb603","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 17:18:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ede9919f28204856b0091ed1e4491181 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/307ab44d1fb049e681dfe4ade952d67c : SUCCESS in 13m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94cad260fc82440bb812243bd052b4c5 : SUCCESS in 13m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65542b554d3c4942b4b7a5292fac6625 : SUCCESS in 12m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/1e4d3b27374849b3a15eaf5464e80ac1 : SUCCESS in 1h 00m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5839af18160040c49b43d428b25322d9 : SUCCESS in 1h 15m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39b18bfcbd0a41e5ab87931673e39533 : SUCCESS in 50m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4785eec8adb497f88fec29ae5d54ad8 : SUCCESS in 9m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/672d6f46dad747e383823acec102462a : SUCCESS in 1h 18m 33s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/56ff8c08c5a9439cad90187c9865dcf4 : SUCCESS in 1h 14m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9394f9882ca747cf98a84df3f5602f38 : SUCCESS in 55m 11s","accounts_in_message":[],"_revision_number":4},{"id":"10e2a9cba0224cea537b0c298d2e258b1970688d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 17:18:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"71f73d87758987f007b0f6822ae56fc11adea937","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 17:22:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3353708a96614411be4990e56567a7e5 : SUCCESS in 1m 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34b5ad5bec5b4219bad34c3b1aa098b1 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":4},{"id":"e492fcbaacea512b65447436f48d56cd9a88c0d0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-03 09:09:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/05c6e2d653dd436e985198ad844c57e8 : SUCCESS in 1h 30m 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b99e73e96cb6448f81a2f4b2e8e624bf36bbff0e","revisions":{"619490a577038f0ffda78eb7e2d59a4ec7720d3f":{"kind":"REWORK","_number":1,"created":"2022-04-07 11:38:22.000000000","uploader":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"ref":"refs/changes/44/836944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/44/836944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/836944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/836944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/836944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/44/836944/1"}}},"commit":{"parents":[{"commit":"85387cae8b3641d6d193c6550edbfb0de1d8384e","subject":"Merge \"mypy: ceph backup driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85387cae8b3641d6d193c6550edbfb0de1d8384e"}]}],"author":{"name":"haailani","email":"haailani@in.ibm.com","date":"2022-04-07 11:37:14.000000000","tz":0},"committer":{"name":"haailani","email":"haailani@in.ibm.com","date":"2022-04-07 11:37:14.000000000","tz":0},"subject":"[SVf]:Fix retype failure for replication volume-type","message":"[SVf]:Fix retype failure for replication volume-type\n\n[Spectrum Virtualize family] While retyping GM to GM or MM to MM or\nGMCV to GMCV, chfcmap() call is being made which is resulting in retype\noperation failure.\nError: CMMVC7154E The task cannot be completed because the specified\nFlashCopy mapping is controlled by a Metro Mirror or Global Mirror\nrelationship.\n\nchfcmap() should not be called for any FC map which is rc_controlled.\nHence while retyping global mirror to global mirror or\nmetro mirror to metro mirror or GMCV to GMCV volume, this error comes\nup as the FC map of the volume is rc_controlled for these replication\nvolume-types.\n\nCloses-Bug: #1968159\nChange-Id: Ic97a2501b4c895b2b655284bcc320036c2d4e217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/619490a577038f0ffda78eb7e2d59a4ec7720d3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/619490a577038f0ffda78eb7e2d59a4ec7720d3f"}]},"branch":"refs/heads/master"},"aced4607478d2aa3a63ee4f0dc37f3bb4501c867":{"kind":"REWORK","_number":2,"created":"2022-04-07 12:11:51.000000000","uploader":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"ref":"refs/changes/44/836944/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/44/836944/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/836944/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/836944/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/836944/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/44/836944/2"}}},"commit":{"parents":[{"commit":"85387cae8b3641d6d193c6550edbfb0de1d8384e","subject":"Merge \"mypy: ceph backup driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85387cae8b3641d6d193c6550edbfb0de1d8384e"}]}],"author":{"name":"haailani","email":"haailani@in.ibm.com","date":"2022-04-07 11:37:14.000000000","tz":0},"committer":{"name":"haailani","email":"haailani@in.ibm.com","date":"2022-04-07 12:10:42.000000000","tz":0},"subject":"[SVf]:Fix retype failure for replication volume-type","message":"[SVf]:Fix retype failure for replication volume-type\n\n[Spectrum Virtualize family] While retyping GM to GM or MM to MM or\nGMCV to GMCV, chfcmap() call is being made which is resulting in retype\noperation failure.\nError: CMMVC7154E The task cannot be completed because the specified\nFlashCopy mapping is controlled by a Metro Mirror or Global Mirror\nrelationship.\n\nchfcmap() should not be called for any FC map which is rc_controlled.\nHence while retyping global mirror to global mirror or\nmetro mirror to metro mirror or GMCV to GMCV volume, this error comes\nup as the FC map of the volume is rc_controlled for these replication\nvolume-types.\n\nCloses-Bug: #1968159\nChange-Id: Ic97a2501b4c895b2b655284bcc320036c2d4e217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aced4607478d2aa3a63ee4f0dc37f3bb4501c867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aced4607478d2aa3a63ee4f0dc37f3bb4501c867"}]},"branch":"refs/heads/master"},"cd80eedad36ff76e5b975cb717a3d3013ad94fd8":{"kind":"REWORK","_number":3,"created":"2022-04-11 11:21:02.000000000","uploader":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"ref":"refs/changes/44/836944/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/44/836944/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/836944/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/836944/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/836944/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/44/836944/3"}}},"commit":{"parents":[{"commit":"85387cae8b3641d6d193c6550edbfb0de1d8384e","subject":"Merge \"mypy: ceph backup driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85387cae8b3641d6d193c6550edbfb0de1d8384e"}]}],"author":{"name":"haailani","email":"haailani@in.ibm.com","date":"2022-04-07 11:37:14.000000000","tz":0},"committer":{"name":"haailani","email":"haailani@in.ibm.com","date":"2022-04-11 11:18:23.000000000","tz":0},"subject":"[SVf]:Fix retype failure for replication volume-type","message":"[SVf]:Fix retype failure for replication volume-type\n\n[Spectrum Virtualize family] While retyping GM to GM or MM to MM or\nGMCV to GMCV, chfcmap() call is being made which is resulting in retype\noperation failure.\nError: CMMVC7154E The task cannot be completed because the specified\nFlashCopy mapping is controlled by a Metro Mirror or Global Mirror\nrelationship.\n\nchfcmap() should not be called for any FC map which is rc_controlled.\nHence while retyping global mirror to global mirror or\nmetro mirror to metro mirror or GMCV to GMCV volume, this error comes\nup as the FC map of the volume is rc_controlled for these replication\nvolume-types.\n\nCloses-Bug: #1968159\nChange-Id: Ic97a2501b4c895b2b655284bcc320036c2d4e217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd80eedad36ff76e5b975cb717a3d3013ad94fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cd80eedad36ff76e5b975cb717a3d3013ad94fd8"}]},"branch":"refs/heads/master"},"b99e73e96cb6448f81a2f4b2e8e624bf36bbff0e":{"kind":"REWORK","_number":4,"created":"2022-04-19 05:20:51.000000000","uploader":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"ref":"refs/changes/44/836944/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/44/836944/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/836944/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/836944/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/836944/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/44/836944/4"}}},"commit":{"parents":[{"commit":"85387cae8b3641d6d193c6550edbfb0de1d8384e","subject":"Merge \"mypy: ceph backup driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85387cae8b3641d6d193c6550edbfb0de1d8384e"}]}],"author":{"name":"haailani","email":"haailani@in.ibm.com","date":"2022-04-07 11:37:14.000000000","tz":0},"committer":{"name":"Harsh Ailani","email":"haailani@in.ibm.com","date":"2022-04-19 05:16:24.000000000","tz":0},"subject":"[SVf]:Fix retype failure for replication volume-type","message":"[SVf]:Fix retype failure for replication volume-type\n\n[Spectrum Virtualize family] While retyping GM to GM or MM to MM or\nGMCV to GMCV, chfcmap() call is being made which is resulting in retype\noperation failure.\nError: CMMVC7154E The task cannot be completed because the specified\nFlashCopy mapping is controlled by a Metro Mirror or Global Mirror\nrelationship.\n\nchfcmap() should not be called for any FC map which is rc_controlled.\nHence while retyping global mirror to global mirror or\nmetro mirror to metro mirror or GMCV to GMCV volume, this error comes\nup as the FC map of the volume is rc_controlled for these replication\nvolume-types.\n\nCloses-Bug: #1968159\nChange-Id: Ic97a2501b4c895b2b655284bcc320036c2d4e217\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b99e73e96cb6448f81a2f4b2e8e624bf36bbff0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b99e73e96cb6448f81a2f4b2e8e624bf36bbff0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
