)]}'
{"id":"openstack%2Fcinder~837311","triplet_id":"openstack%2Fcinder~stable%2Ftrain~Ibb77b8b1be6c35c5db3b07fdc4056afd51d48782","project":"openstack/cinder","branch":"stable/train","topic":"img-fmt-ussuri-stable/train","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-04-22 06:17:15.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2022-04-22 06:17:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibb77b8b1be6c35c5db3b07fdc4056afd51d48782","subject":"Reject bad img formats for uploaded encrypted vols","status":"MERGED","created":"2022-04-11 15:41:06.000000000","updated":"2022-04-22 06:18:43.000000000","submitted":"2022-04-22 06:17:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837311-img-fmt-ussuri-stable/train","cherry_pick_of_change":828890,"cherry_pick_of_patch_set":1,"meta_rev_id":"17c057e059f4ddae4427884accbb138ed73c39df","_number":837311,"virtual_id_number":837311,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-22 06:17:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-19 12:32:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2022-04-14 15:04:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-19 12:32:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-11 18:47:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-11 20:20:54.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-11 20:32:09.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-04-11 23:51:40.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-12 03:27:27.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-04-12 09:09:01.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-04-14 15:01:30.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2022-04-14 15:04:59.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-04-19 12:32:02.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"e3aaa49be72cd08e6f459c07d8532f213d97b8bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-04-11 15:41:06.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"7c42d27cc6504ec181865d9dad5da039168d5f41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 18:47:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/785fd7494ab342eca9504223deadde5a : SUCCESS in 8m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cec4e94165614245881e9218a1d5bb4b : SUCCESS in 9m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a32e6bfdd04e4a7ca54467815afbc3aa : SUCCESS in 9m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33182f5580f64281b15da07b6f911282 : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54d1207e291a4a10a2762983fd730124 : SUCCESS in 13m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7e3377f6018b45f7b5e2f0b2842db6a5 : SUCCESS in 1h 29m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3f9c40a013d44145837eb878d565f708 : SUCCESS in 1h 04m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/e5b425d157694afa94320bdeb3ba78b0 : SUCCESS in 59m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ffab4c677dfc46fea774dce443d944fc : FAILURE in 22m 55s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/635284dba565457689f41573601c6adf : SUCCESS in 1h 05m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dedcc45641144d1fb9b8431bb4d3cb70 : SUCCESS in 48m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7103225ed26640ff8fd66eaef7f4cf02 : SUCCESS in 7m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bfcc21b059f748d8a007dc0a5fbcb085 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3cfdb30cb5394934b0edd409c9ef0ce0 : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/45d637f48df04f5a8634fbcd6ae95221 : FAILURE in 23m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7d4cbe3069364b388b6bba9a6d92d9fd : SUCCESS in 6m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c111c88c5698485aaac660d8c202ed4e : SUCCESS in 1h 24m 13s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/055ca7d92fd34cd1bbda5a220d005c82 : SUCCESS in 6m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e9686bb0e2c34114b4f1e629197c6772 : SUCCESS in 1h 43m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1be35cdcad03451eb59bb8382ab6156a : SUCCESS in 1h 25m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8e8a1b409dfa4fb0bdbf82041bb6becc : FAILURE in 1h 46m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/eaba3c50cd20413bbe4d652ff5c43ffb : SUCCESS in 1h 31m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7deadcbb34ef493792d06c7e00f508ce : FAILURE in 2h 06m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d9bae3bfba8b4cccaf5caede3fbaa2d0 : SUCCESS in 1h 02m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1e0503fd55a43d0a8b3172b0cb237b0 : SUCCESS in 59m 41s","accounts_in_message":[],"_revision_number":1},{"id":"65d9a27e826cdbc5fc42e9d481736ea17dd0ea59","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-11 20:20:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/837311 : FAILURE in 51m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1f07f878005e506e55df36f0dbcef3074efaeae1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-11 20:32:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/11/837311/1/check/iscsi-driver-alletra-6k/78b3bf9/78b3bf9.html : FAILURE in 53m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5848b1e652f4bf8c0d709f8b24c3fc336ef955c2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-11 21:43:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/837311 : FAILURE in 54m 40s","accounts_in_message":[],"_revision_number":1},{"id":"5073d7e85e4b1161be7f30e306dfafafba9eb7dc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-04-11 22:33:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"452ce0366d539bc8b5a874bbeb86c4712fe1c8f3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-11 23:51:40.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/51574a999c774df0a91e75190683b38c : TIMED_OUT in 3h 07m 53s\n- devstack-plugin-ceph-dsvm-bionic-py3 finger://zuul-v3-executor.novalocal/d176eed52d7c4cfabfc00d7f24c7f587 : TIMED_OUT in 3h 08m 57s","accounts_in_message":[],"_revision_number":1},{"id":"db984bf2f00d720e09eb501cd6bed3570cc010e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 00:29:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/107589fbdf6b455b8e2052c0b14d5ff1 : SUCCESS in 8m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4d00c9a086ef402d8fe184f61e07fcd2 : SUCCESS in 8m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e3fac5133a9481d945b11d0fecd9394 : SUCCESS in 9m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae50df1b827b465b8e8320e4b1dc6abd : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96c2f5d847b94f9a8449d6bc774af3b9 : SUCCESS in 13m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/76e1062e364e4402b0b8df21a153dceb : SUCCESS in 1h 27m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/967ed248d68244cba48431b41f7551b6 : SUCCESS in 57m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/5b574b3726cf4c0b9728d9aaa935ac20 : SUCCESS in 56m 36s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4dcc6bf49fd543c0b702e7a32f49745f : FAILURE in 15m 26s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/61386b6ac1af45248efe4044a1c5141d : SUCCESS in 1h 25m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e247a1c9bf3649f48ad5936d0b829826 : SUCCESS in 52m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74f7ba40d4fc40a8a5b190b0dc454abe : SUCCESS in 8m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b7a2f8cde61f4e6090d5fd64f69debca : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e383ddb2a94f4760967d3a382b66c780 : SUCCESS in 6m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e325d398284c41c5baa4c46e58d53763 : FAILURE in 24m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/418cb8efcece4965b34abb470ce22afe : SUCCESS in 7m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/11d3e693ec9647bb9ea3eec9674b676a : SUCCESS in 1h 14m 29s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/29ac65c3c4af4f3c9f7b257c06b31b8c : SUCCESS in 6m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c1fbb2f0f18d4181b746215fa5fd8f7a : SUCCESS in 1h 24m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3c4b4b3d847a450ab9c81b571bff6b76 : SUCCESS in 1h 19m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/de65d9ba615047088ba19dc1552b8317 : FAILURE in 1h 17m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b299d48194ba4b0ea10dccbbe3f7fdbe : SUCCESS in 1h 49m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7eca0a768aba4f8ca2e2f686fbb91842 : SUCCESS in 1h 42m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8c98e9c637fd4d5bba4a7432dca17eed : SUCCESS in 1h 05m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74aaa629ad9b44049b2c319e2a5b0831 : SUCCESS in 1h 17m 11s","accounts_in_message":[],"_revision_number":1},{"id":"36fb91e9504838955d40ac7dd42a947d77106dd3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-12 03:27:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/11/837311/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8ecc472/ : RETRY_LIMIT in 3m 45s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/11/837311/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/433ca90/ : RETRY_LIMIT in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"12d7455b5262d27b7bb7b313ff1a90af9031b96c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-12 06:45:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"dbb6b2297a7e415cfd00edf60ffd4ff6978a4d68","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-12 06:45:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"6c9643391dba53c2fd4ac231b2f31396b646cb97","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-12 09:09:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/11/837311/1/check/dsvm-tempest-cisco-zonemanager-job/c32f2fd/ : FAILURE in 21m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1a38f710c1f6b4d46fa525506dc87488482ab8ac","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-04-14 15:04:59.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a7cdd10716a2947d4bcd746ca2097c05f20c2b2f","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-04-19 12:32:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6848ccb4df9e6fae29d017ab7bcdfe24ca3bb71b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 12:34:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ecfadd9718ac57562d62b21971147dce9365aa98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 13:04:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c44326b837804c9ca6b7392e5d7e1139 : SUCCESS in 9m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/72a72e42d56749a7bb4030284a27d508 : SUCCESS in 9m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7772f737e29a4ac880ff9529e797ae20 : SUCCESS in 11m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/36869f83c7324b72a2ef6e8738df483c : SUCCESS in 12m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea3d4dc7f54240809c9aea7a5a65821a : SUCCESS in 15m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/53431c097e864a069611f8e1b8533ea6 : FAILURE in 18m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d34ad2c7f3584fed8031030246877de2 : FAILURE in 24m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/8a870a4954574f0ea90453ae437e17c8 : FAILURE in 18m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b136b2b5d8b246d29f05bbe9c73b12a6 : FAILURE in 17m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88b9444a4bd749a8bc89dc0ea87a4a4b : POST_FAILURE in 18m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b19f64c4b3c49f787b3ada542ac9113 : SUCCESS in 9m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/76fa3b98e3184efe9bd327fad34b2c1b : FAILURE in 25m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/60d2c1c0c2564e3985c9ec723d6732ca : FAILURE in 18m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4db930e448314b5397368fda473d2943 : FAILURE in 17m 12s","accounts_in_message":[],"_revision_number":1},{"id":"4a64350e868563772a17855d2d347398eb90638f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-04-19 16:22:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"082f8a6c89bf0fec2d2774c02946de4ae6ae426c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-04-19 17:08:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5730570a67763c1efab4737a4603d746f5e8d869","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-04-22 02:17:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5f8346cf0c5bb0315e9cd93bf10b6fa74efd4a5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-22 04:11:43.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/11/837311/1/check/tempest-dsvm-full-bionic-py3/a62afb5/ : SUCCESS in 1h 47m 55s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/11/837311/1/check/devstack-plugin-ceph-dsvm-bionic-py3/3882e08/ : SUCCESS in 1h 50m 23s","accounts_in_message":[],"_revision_number":1},{"id":"361e2ede8a41483f88c691ed259639c2f3dafcbc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-22 04:24:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : RETRY_LIMIT in 0s","accounts_in_message":[],"_revision_number":1},{"id":"1b9a522091b43cd6f186b3dde01ff3d92c03a0c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 04:35:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68c52a4f00f344ad8eca3353388b0aa4 : SUCCESS in 8m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7dbc3cac562e4921a954cb854df26176 : SUCCESS in 11m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c7bcbf97f02428d861bf7dc1815438e : SUCCESS in 11m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ee62844b895428ba95c758088e16c69 : SUCCESS in 10m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18211eef4a9c4bbf84f0d4a257223ac1 : SUCCESS in 13m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1a08da0def4843b2bdc2f2098eab6497 : SUCCESS in 1h 27m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/98384a7ca8384e9e999c305e4afd8fdc : SUCCESS in 1h 07m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/3ffeca7a3d6e484c9c15eb7990261924 : SUCCESS in 1h 02m 16s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/face4fefd2944be4bb86daac5d7866a8 : FAILURE in 17m 16s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b318620464604595b9a684ef8f4ad5b8 : SUCCESS in 1h 07m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56a9765450784abf990ac7f14baca94b : SUCCESS in 52m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd5d70df9a554a5384af086351051081 : SUCCESS in 9m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b9ad3a764d7b45198941d304a361d031 : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/48113114ea19464dbf5fdd0ca1bb044b : SUCCESS in 8m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/523c330412fc4e1a887c54d904c2e37e : FAILURE in 23m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6b113724d5a54cf1ba7fe702b223f18d : SUCCESS in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/404a8035248a4fde9157420da8a99911 : SUCCESS in 1h 20m 33s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/61ca3ee2cd7d4afaa9202291aa47daed : SUCCESS in 7m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/311ebe2d162b4fd9a28b2fe5d9723cbb : SUCCESS in 1h 09m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dcd71dbd4da34ad194be7ff1618b0191 : SUCCESS in 52m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7ed1ba61e21540e98c50ff88ce1b1b5a : FAILURE in 1h 12m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/98ca44f3322b4c088817b12819c989e3 : SUCCESS in 1h 25m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/487eb0b2164a4f7fbaed127ee8cfc869 : SUCCESS in 1h 49m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ef2b6ba7b4b246dda7013528cf9d0096 : SUCCESS in 56m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25349f4dcc84424c93352f99dc5629cc : SUCCESS in 1h 05m 49s","accounts_in_message":[],"_revision_number":1},{"id":"210b8e6a13a5562ddc35a7572e20e26610ddb37c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 04:36:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"829c26881879c0656bb34a4a8098b2d28a5c7fc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 06:17:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/073442c1f7454c8c9d0e4435ea813dfe : SUCCESS in 8m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4fe0173fe12a40cfbad1f107e0340ea3 : SUCCESS in 9m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae3572a1e20c409da00e345ac9bd558c : SUCCESS in 8m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/89afa41c3e8149b4961209037361906f : SUCCESS in 10m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82fbd6ae96834d22ac80c27f6039860b : SUCCESS in 14m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f6e9c4e9ab574f78aaee7df64d357847 : SUCCESS in 1h 34m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/aa031b46965d447d86ac4b6962876fa4 : SUCCESS in 1h 18m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/e5d8625348de47bda92732ea87dc60e2 : SUCCESS in 1h 01m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e78787bfb9264b3f93ce516691406367 : SUCCESS in 1h 10m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39e16d7dd19c47e19a983e7c6811c181 : SUCCESS in 55m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a284a65c19714cebbe1abbb919d43b2a : SUCCESS in 6m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5d81891a1d5241c58240ef764ef4b340 : SUCCESS in 1h 18m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/66e760f6d2cb4f698162b4c98fcfc19c : SUCCESS in 1h 06m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43a1f2508f1f4439931cdeef1422c1b2 : SUCCESS in 1h 09m 46s","accounts_in_message":[],"_revision_number":1},{"id":"69cf572b462d626a92d48d810fb2c71a414f210c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 06:17:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"17c057e059f4ddae4427884accbb138ed73c39df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 06:18:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b78a47f73c84a1fb8767305bfbff7d0 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e4d42adfd9c464181a06dd1b86d99b8 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"513ba07ed6a434313f4f6e66d113b4368d4694a7","revisions":{"513ba07ed6a434313f4f6e66d113b4368d4694a7":{"kind":"REWORK","_number":1,"created":"2022-04-11 15:41:06.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/11/837311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/837311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/837311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/837311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/837311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/837311/1"}}},"commit":{"parents":[{"commit":"c093edab2d11938ac88cbde3323b90d50e8120a6","subject":"Merge \"RBD: Open RBD images read-only where possible\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c093edab2d11938ac88cbde3323b90d50e8120a6"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-07-09 15:08:34.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-04-11 15:41:06.000000000","tz":0},"subject":"Reject bad img formats for uploaded encrypted vols","message":"Reject bad img formats for uploaded encrypted vols\n\nCinder only supports uploading volumes of encrypted volume types as\nimages with disk format \u0027raw\u0027 and container format \u0027bare\u0027.  Screen\nfor this at the REST API layer when the request is made.\n\nThis change is applied directly to the VolumeActionsController in\nthe cinder.api.contrib module, so it affects both the Block Storage\nAPI v2 and v3.\n\nChange-Id: Ibb77b8b1be6c35c5db3b07fdc4056afd51d48782\nCloses-bug: #1935688\n(cherry picked from commit de8b3b0b00421a01ed777880d2b5bf6fffc068d5)\n(cherry picked from commit 78682022d2dcf07c270356d23a27843578466469)\n(cherry picked from commit 12f7376fe44ea53e6e84dfd12577953bd841d25d)\n(cherry picked from commit fac676112da484e8ad2627de4333ab72b70fe5c8)\n(cherry picked from commit 7c3f7379feb41664c3881de99b54ca13a6b8ec61)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/513ba07ed6a434313f4f6e66d113b4368d4694a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/513ba07ed6a434313f4f6e66d113b4368d4694a7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
