)]}'
{"id":"openstack%2Fcinder~837662","triplet_id":"openstack%2Fcinder~master~I6ea918e31e4c6a272e6e4ba86a7bcc5d9a0ac3f4","project":"openstack/cinder","branch":"master","topic":"hw-drv-stability","attention_set":{},"removed_from_attention_set":{"10487":{"account":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"last_update":"2023-06-20 10:19:29.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6ea918e31e4c6a272e6e4ba86a7bcc5d9a0ac3f4","subject":"Update huawei driver","status":"ABANDONED","created":"2022-04-13 06:22:41.000000000","updated":"2023-06-20 10:19:29.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f2dbecf84a99f3114b291798d96b67f26dc1efe8","_number":837662,"virtual_id_number":837662,"owner":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-13 06:53:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-13 06:22:52.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-13 06:24:02.000000000","updated_by":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REMOVED"},{"updated":"2022-04-13 06:24:31.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-13 06:28:15.000000000","updated_by":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"reviewer":{"_account_id":20284,"name":"huawei-cinder-ci","email":"liuxinguo@huawei.com","username":"huawei-cinder-ci"},"state":"CC"},{"updated":"2022-04-13 06:28:16.000000000","updated_by":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REMOVED"},{"updated":"2022-04-13 06:30:26.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-13 06:41:52.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-13 06:43:17.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-04-13 06:53:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-13 07:03:13.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-13 07:04:14.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-04-13 07:07:53.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-04-13 07:26:48.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-14 03:47:30.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-04-18 04:20:01.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-04-19 12:13:24.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2022-04-22 13:18:08.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-02 13:50:18.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"9ea976b775ffda2427709aee0ac1d3824934232a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"date":"2022-04-13 06:22:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8d96c952568abfba076223158266355ab986ad2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-13 06:22:52.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15821/","accounts_in_message":[],"_revision_number":1},{"id":"aae27f1df7b6a7ad9902310abfebf195edd311a8","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"date":"2022-04-13 06:24:02.000000000","message":"Removed cc VMware NSX CI.","accounts_in_message":[],"_revision_number":1},{"id":"c468f5334feafda6d4c3b20ec70fcf64c9547b44","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-13 06:24:31.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15821/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"90145249e340a67ea09bce91abaa11975e9864d0","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"date":"2022-04-13 06:28:16.000000000","message":"Removed cc VMware NSX CI.","accounts_in_message":[],"_revision_number":1},{"id":"d716b4af8a4279e10fd31e7d22f1b4fac89d4d40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"date":"2022-04-13 06:30:15.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8014ce886612d9de4f733bfaca4cd1ce8cea1eed","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-13 06:30:26.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15823/","accounts_in_message":[],"_revision_number":2},{"id":"ca91dd4faacf107a2fff2b99c58045e726178530","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-13 06:32:10.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15823/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"4f5d4fa2ef6165f8ec72e63aa2fa2968df2e08ff","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-13 06:38:58.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/837662/837662/2/check-not-vote/ext-cinder-zuul/7da350b : FAILURE in 8m 01s","accounts_in_message":[],"_revision_number":2},{"id":"091ac242d476bf96594c5405593bb0e1db9b015c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-13 06:41:52.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-62-837662-1/ns5_nfs : FAILURE in 15m 52s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-62-837662-1/ns5_iscsi : FAILURE in 18m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"36f8d7c283c55ecdc52e6aa1f1eff0dd65f47ea1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-13 06:43:17.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/62/837662/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/cfe6d02/ : FAILURE in 11m 22s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/62/837662/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ddedbbf/ : FAILURE in 11m 12s","accounts_in_message":[],"_revision_number":2},{"id":"4a52ccaa14077baa22980299ad38d535580c3f47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 06:53:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01c0f03b833a449b87ceec1693d08375 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a94526349f2e4fefa809a385c2550fd4 : SUCCESS in 10m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c96db8fa318047d8bc7c2e2e2c692ead : SUCCESS in 11m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94267bb5a0304f8b8bfac1a6088c811b : RETRY_LIMIT in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a6bef556e6543de9a29b1f401360c21 : SUCCESS in 10m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/e32450712d274e80895a65c5c3bc9a06 : FAILURE in 10m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/43672c8489484a189bf4412e2dea4e6d : FAILURE in 10m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9942e7318f54a09b87fb4e0ad3e7082 : POST_FAILURE in 12m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/298c95d52b4f478d888078fb9559e763 : SUCCESS in 14m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c1eeedab3f12491ca0ec52c0f29b0b32 : SUCCESS in 6m 12s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6dd74db0bd234adba90d2bdb2360aca4 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/99973e050a23416d8d96fd7ca701d7e4 : SUCCESS in 6m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/31cf3e4af8b341b88968e75b23ee0626 : SUCCESS in 8m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3a8463e971fe4e9d978d5be61dc9eaba : FAILURE in 11m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/afca306c80ac4ad592af38dc96bfb017 : FAILURE in 4m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/791da63009974d7590ba5ad46bc11a40 : POST_FAILURE in 13m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/08c0e561d4ca45d2a0d5e38e55553900 : POST_FAILURE in 16m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dc376b0dd182447192a0efde5d962724 : POST_FAILURE in 13m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/23a9a8df5ac04574ad6f4ea7c0bcdbc3 : FAILURE in 21m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/436e87de6b7345fb9d6456a9c4aaab52 : FAILURE in 11m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/793cade762e54f3da658eb0f3e9972a2 : FAILURE in 11m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f4111e1bd1c848389bc0631271aaaafd : FAILURE in 13m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/66ea2e52d0e14885a69fb2a9d52e3e00 : FAILURE in 14m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fb2224d5ff2499aac0b64cc09162c62 : FAILURE in 11m 33s","accounts_in_message":[],"_revision_number":2},{"id":"166aa33334aad726839d37cd97e3b9861f6fc37e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-13 07:03:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/837662 : FAILURE in 31m 18s","accounts_in_message":[],"_revision_number":2},{"id":"eb577efe62e238c6e25f84707fd39a5c50268de2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-13 07:03:24.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-62-837662-2/ns5_nfs : FAILURE in 14m 9s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-62-837662-2/ns5_iscsi : FAILURE in 17m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"a36532f4572dd427d3fecda9012fe1bc7e26f5d3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-13 07:04:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/62/837662/2/check/iscsi-driver-alletra-6k/e5502fe/e5502fe.html : FAILURE in 30m 48s","accounts_in_message":[],"_revision_number":2},{"id":"51b839269b5e9f9fe53c8da900ca69b9f3e8a4ed","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-13 07:04:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/837662 : FAILURE in 32m 58s","accounts_in_message":[],"_revision_number":2},{"id":"45bb5854e46278bb0c0cff9a1ffc2bdf39a60a8e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-13 07:07:53.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/837662/2/iscsi/ : FAILURE in 33m 08s\n- cinder-smb http://cloudbase-ci.com/cinder/837662/2/smb/ : FAILURE in 37m 08s","accounts_in_message":[],"_revision_number":2},{"id":"07c49bb3656c45804a11aa2778f82e0c51790484","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 07:09:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ffb571f3e9384b6d919fa539ee2e6bdf : FAILURE in 36m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/515998b8d757440fa2d58b9142279167 : SUCCESS in 20m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf4de405c3b907967b19822c4680c588b724dc97","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-13 07:26:48.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/2f6611413d384a068b32d1016173adab : POST_FAILURE in 51m 20s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/19bab6ad07dc422c9b5afa0f71e185e1 : POST_FAILURE in 47m 02s","accounts_in_message":[],"_revision_number":2},{"id":"e73cd50d78e23dc1fb308faee98087bd2a7c577e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-04-13 19:21:36.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-62-837662-2/ns5_nfs : SUCCESS in 3h 22m 26s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-62-837662-2/ns5_iscsi : SUCCESS in 3h 23m 7s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"1828ec95e621c8c5deb96e67886e2f9e49429ffa","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-14 03:47:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/62/837662/2/check/dsvm-tempest-cisco-zonemanager-job/a29eadd/ : SUCCESS in 1h 13m 26s","accounts_in_message":[],"_revision_number":2},{"id":"5307fba27404ef507f0a57b78ded8c4a5e8a2665","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-14 07:39:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/62/837662/2/check/iscsi-driver-nimble/5acf97f/5acf97f.html : FAILURE in 1h 35m 50s","accounts_in_message":[],"_revision_number":2},{"id":"95dd90bc9823de9591fee771451eb71a924aa344","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-14 09:20:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/62/837662/2/check/iscsi-driver-nimble/b76a398/b76a398.html : FAILURE in 1h 38m 47s","accounts_in_message":[],"_revision_number":2},{"id":"514c06d7ea932d971778161819b3ba857aee535e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-04-18 04:20:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/837662/2/20220418104909/NEC-V-FC : SUCCESS in 1h 53m 30s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/837662/2/20220418104909/NEC-V-ISCSI : SUCCESS in 2h 30m 48s","accounts_in_message":[],"_revision_number":2},{"id":"414e5a117c103e3cc897df50a83eab2a39a847b7","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-04-19 12:13:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"facf6a8970c5a6abdef8688a406664068dd456f2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-19 12:44:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/837662 : FAILURE in 29m 22s","accounts_in_message":[],"_revision_number":2},{"id":"2de0c416230cc2661dc007ad65c83f73636dd6df","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-04-22 13:18:08.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/837662/2/20220422155757/NEC-FC : SUCCESS in 3h 18m 39s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/837662/2/20220422155757/NEC-ISCSI : SUCCESS in 6h 20m 07s","accounts_in_message":[],"_revision_number":2},{"id":"0ab60fd06a5ee9dab0aec4e7f9ce56cf336d8e1d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-04-25 09:10:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"98e8bd63282e79e80456a618bd3d006c0fa76797","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-25 10:45:16.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/62/837662/2/check/iscsi-driver-alletra-6k/f754bf7/f754bf7.html : FAILURE in 1h 34m 06s","accounts_in_message":[],"_revision_number":2},{"id":"deb94f137be80a088e6ebd51dc029ff64443dfbb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-02 13:50:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3fa92e3171ff44a3b990d52d3d56de83 : FAILURE in 19m 25s","accounts_in_message":[],"_revision_number":2},{"id":"f2dbecf84a99f3114b291798d96b67f26dc1efe8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"date":"2023-06-20 10:19:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6e17412a2dcedeea716d3333e0b45b39da6b9952","revisions":{"e738748f40c6e8434f1014ab19f79b17933ee1c9":{"kind":"REWORK","_number":1,"created":"2022-04-13 06:22:41.000000000","uploader":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"ref":"refs/changes/62/837662/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/837662/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/837662/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/837662/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/837662/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/837662/1"}}},"commit":{"parents":[{"commit":"6d4a6aa978efd094c25dfd14239e40564c0227d7","subject":"Merge \"Remove privsep/hscli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d4a6aa978efd094c25dfd14239e40564c0227d7"}]}],"author":{"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","date":"2022-04-13 04:58:45.000000000","tz":330},"committer":{"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","date":"2022-04-13 06:21:55.000000000","tz":330},"subject":"Update huawei driver","message":"Update huawei driver\n\nImprove the huawei driver code test coverage\n\nChange-Id: I6ea918e31e4c6a272e6e4ba86a7bcc5d9a0ac3f4\nSigned-off-by: Kanagaraj Manickam \u003ckanagaraj.manickam@huawei.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e738748f40c6e8434f1014ab19f79b17933ee1c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e738748f40c6e8434f1014ab19f79b17933ee1c9"}]},"branch":"refs/heads/master"},"6e17412a2dcedeea716d3333e0b45b39da6b9952":{"kind":"NO_CHANGE","_number":2,"created":"2022-04-13 06:30:15.000000000","uploader":{"_account_id":10487,"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","username":"kanagarajm"},"ref":"refs/changes/62/837662/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/837662/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/837662/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/837662/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/837662/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/837662/2"}}},"commit":{"parents":[{"commit":"6d4a6aa978efd094c25dfd14239e40564c0227d7","subject":"Merge \"Remove privsep/hscli\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d4a6aa978efd094c25dfd14239e40564c0227d7"}]}],"author":{"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","date":"2022-04-13 04:58:45.000000000","tz":330},"committer":{"name":"Kanagaraj Manickam","email":"mkr1481@gmail.com","date":"2022-04-13 06:29:42.000000000","tz":330},"subject":"Update huawei driver","message":"Update huawei driver\n\nImprove the huawei driver code test coverage\n\nChange-Id: I6ea918e31e4c6a272e6e4ba86a7bcc5d9a0ac3f4\nSigned-off-by: Kanagaraj Manickam \u003ckanagaraj.manickam@huawei.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e17412a2dcedeea716d3333e0b45b39da6b9952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6e17412a2dcedeea716d3333e0b45b39da6b9952"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
