)]}'
{"id":"openstack%2Fcinder~839469","triplet_id":"openstack%2Fcinder~master~I1e0d0a43cd166e6cfd4be7ca0f426e9fc3625a28","project":"openstack/cinder","branch":"master","topic":"drop-c8s-testing","attention_set":{},"removed_from_attention_set":{"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2022-04-28 11:23:42.000000000","reason":"Luigi Toscano replied on the change"},"8846":{"account":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"last_update":"2022-04-29 19:50:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1e0d0a43cd166e6cfd4be7ca0f426e9fc3625a28","subject":"Drop python3.6 support in testing runtime","status":"MERGED","created":"2022-04-27 06:14:46.000000000","updated":"2022-05-04 16:53:12.000000000","submitted":"2022-04-29 19:50:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839469-drop-c8s-testing","meta_rev_id":"41a35d2172c3b3087aa527392cd12edec94aa86b","_number":839469,"virtual_id_number":839469,"owner":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 19:50:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-04-29 14:37:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-04-28 11:23:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-29 14:28:53.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2022-04-29 14:28:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-04-29 18:24:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2022-04-29 18:24:10.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 06:14:56.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-04-27 06:14:57.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-27 06:43:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-27 06:48:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-27 07:55:14.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-04-27 08:13:35.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-27 08:30:08.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-04-27 08:58:29.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-27 09:28:56.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-04-27 09:35:05.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-04-27 09:58:14.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-04-27 10:19:18.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-27 10:31:12.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-04-27 10:55:28.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-04-27 11:15:26.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-27 11:15:53.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-04-27 11:53:20.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-04-27 11:56:04.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-04-27 12:42:44.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-04-27 14:20:22.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2022-04-28 04:26:11.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-28 04:26:29.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-04-29 09:10:55.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-04-29 11:15:15.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-04-29 14:28:53.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2022-04-29 14:28:57.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2022-04-29 14:37:09.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-04-29 15:30:45.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-02 05:20:41.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-05-04 16:53:12.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"cef518185eb28ffd364439bdfde112453dc10d59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2022-04-27 06:14:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7a15ac5d5d6ce95ff6a7ab0f5c84d0aedc85e5e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-27 06:14:56.000000000","message":"Patch Set 1:\n\nBuild Started","accounts_in_message":[],"_revision_number":1},{"id":"8709094eabf4821a01e865fc882234e2f85005fe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-27 06:14:57.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15892/","accounts_in_message":[],"_revision_number":1},{"id":"0744f573150385a3670d8986352adac6ed09eac9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-27 06:16:44.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15892/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"fd70c9839f080413d467a7b7f73b8b8c0a6c1382","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-27 06:32:12.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1651040095551 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"88dc079278cef82fa43875d6bfd99524eac52ba3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 06:43:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe264dc5ad6f4db7b8a77815d03dcf5a : SUCCESS in 24m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b3382e037cf84fe9a27ce81100a4687a : SUCCESS in 26m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d205d05e617d3459b5e06f6dd4017467834cc8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 06:48:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e3e15040d5974b4c8c95ba0d62e5b64b : SUCCESS in 2m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d68c848fa4d24b4189581c7eed872d74 : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85292249337d41a09102c397a716ae1f : SUCCESS in 9m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74d3a3487ad84bf7abc2570de8382c21 : SUCCESS in 9m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7b2fd0e96bc476ba0e14c04659f7058 : RETRY_LIMIT in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48e7990e1f274bda8351efa9da1750bb : SUCCESS in 10m 07s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c48e469503a14791bd0461eb513328b5 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/864124a005c0454cb8020348512569d3 : FAILURE in 4m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a206b08d59bb4bd988226f12e1836ae1 : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/94f9ad96518f44bbadc313cf8e1dcf2b : SUCCESS in 31m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/07f22b10cb6844598dcb7373ac466cd7 : SUCCESS in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d5ed070d5edc4b6fe4195b93626fedcc7b6427c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2022-04-27 06:53:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a777cae3f57a9cacbec1125bd313115cba742871","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-27 06:53:22.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15893/","accounts_in_message":[],"_revision_number":2},{"id":"c909a07f2819d1eac3a54ff273e9c63abbcfa740","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-27 06:53:26.000000000","message":"Patch Set 2:\n\nBuild Started","accounts_in_message":[],"_revision_number":2},{"id":"fc3bfc11d45e010fab6587f83166c14213f1ddd8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-27 06:55:12.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15893/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"d0b996790a95b5ea1cb1f7df45b88c70069a64f0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-27 07:10:42.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1651042405618 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"48994fc7aab55fed0511ce1382ccd8bfd07da7e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 07:22:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9119a8273e804c67b9afc0847099e796 : SUCCESS in 27m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/84fcfd5c508746499031af1323d3f997 : SUCCESS in 24m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"53244cdf76768aba6d27f9d11e799e2424ae018a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-04-27 07:55:14.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/839469/2/iscsi/ : FAILURE in 1h 00m 56s\n- cinder-smb http://cloudbase-ci.com/cinder/839469/2/smb/ : FAILURE in 50m 11s","accounts_in_message":[],"_revision_number":2},{"id":"e9f2c18c5f9b7402d3ecde32b0df8e55963d827f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-27 08:13:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/839469 : SUCCESS in 1h 17m 58s","accounts_in_message":[],"_revision_number":2},{"id":"0e19b38ee0a4484229b6935aeaf867da03236cff","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-27 08:30:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/69/839469/2/check/iscsi-driver-alletra-6k/d843eb1/d843eb1.html : FAILURE in 1h 35m 39s","accounts_in_message":[],"_revision_number":2},{"id":"e1f6bd2ea55f9e70487772ced370074d3c8d2fed","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-27 08:30:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/69/839469/2/check/iscsi-driver-nimble/420ff33/420ff33.html : FAILURE in 1h 35m 48s","accounts_in_message":[],"_revision_number":2},{"id":"67c80417d4845db1293ebd7769768ca298943fc0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-27 08:58:29.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/69/839469/2/check/tempest-dsvm-full-focal-py3/55300cc/ : SUCCESS in 2h 02m 42s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/69/839469/2/check/devstack-plugin-ceph-dsvm-focal-py3/6bbd1a0/ : FAILURE in 30m 11s","accounts_in_message":[],"_revision_number":2},{"id":"de271dc9cc8b78530031ce5fedb8c4ffa8fd459e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 09:09:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/481ccded8a534b2489391bb9d858897f : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dae231bb11546268f0a7ab8e9233010 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16f550d7875a4f4992459312b104a0a4 : SUCCESS in 10m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d082d2cfa2b54cc59aabb2565c8b39b0 : SUCCESS in 11m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8880918301264c9d8e0ffc00117bc6f7 : RETRY_LIMIT in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faccf319a36a48d689e61d1848b04451 : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/8200d62706474566a1ec27d115636992 : SUCCESS in 59m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/73fad17349064ca69b28f97ab7768252 : SUCCESS in 1h 19m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/929467484d6b4cffa05f9a26065e0d01 : SUCCESS in 50m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/922b16788c7c425795f1f0fc2598d41f : SUCCESS in 12m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cbc3443c30244ec684ecdabb1b7738b7 : SUCCESS in 5m 17s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/da718bec0e21412db0f8ab9014092b87 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d7b24c1dd0fb418f9a137ca80c9fd20c : SUCCESS in 6m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/00c651e453624d9f861c6bd45d42f110 : SUCCESS in 6m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9a03ae5ac5fe4c99ad2623fc8be66ad2 : SUCCESS in 44m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ea3a050670fa4ecc90a96bca57512635 : SUCCESS in 5m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e871ecea78bc4c6fb5fa580e92f10bfd : SUCCESS in 1h 14m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/682d72f2e6fc4f99a7031ff5506d4d60 : SUCCESS in 1h 22m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fd304aeeb9e8440c8a8314c834477f32 : SUCCESS in 1h 52m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bcb2cd8a91924c68b415612a1735bbd9 : SUCCESS in 1h 09m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/507cdb687d6847cb9147537f08476d7f : SUCCESS in 1h 25m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bda566c0cef44b59b647c0b407f1b3c5 : SUCCESS in 40m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/00e6be2948a24d31a8658b27f95b29d9 : SUCCESS in 1h 20m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1c37a9450569495aa2a9295650aa07e2 : SUCCESS in 2h 13m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88ccde09d3ae4e129eb7f731d53a278d : SUCCESS in 54m 41s","accounts_in_message":[],"_revision_number":2},{"id":"f2ef0e88533c104fa53c9805a3802457b1f2140c","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-04-27 09:28:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-69-839469-2/1651050595 : SUCCESS in 18m 59s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"e9528e73384725505b887f3cc66f072249bcdca0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-04-27 09:35:05.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/27_04_2022_09_10_02\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"835fa523fc5c618734bf66bd5575e876b579fec0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-27 09:58:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/69/839469/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/737af4b/ : TIMED_OUT in 3h 03m 07s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/69/839469/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/079cc8a/ : SUCCESS in 1h 04m 10s","accounts_in_message":[],"_revision_number":2},{"id":"e7f84c1d0530d1734f1e82e10faa458188bda033","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-04-27 10:19:18.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/69/839469/2/check/DellEMC_PowerFlex/1345bd9/DellEMC_PowerFlex/659 : SUCCESS in 1h 09m 05s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"739227c92aa2a4aba9f980ebfd838970ab250672","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-04-27 10:31:12.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/69/839469/2/check/DellEMC_PowerStore_iSCSI/ee572b4/DellEMC_PowerStore_iSCSI/454 : SUCCESS in 1h 20m 57s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/69/839469/2/check/DellEMC_PowerStore_FC/0b5eab9/DellEMC_PowerStore_FC/469 : SUCCESS in 1h 08m 31s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"eec99e55d16ce0a2a0329c5678b5fc5e0f5c149f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-04-27 10:55:28.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/69/839469/2/check/DellEMC_SC_FC/b850708/DellEMC_SC_FC/531 : SUCCESS in 1h 00m 11s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/69/839469/2/check/DellEMC_SC_iSCSI/155c08d/DellEMC_SC_iSCSI/552 : SUCCESS in 1h 45m 17s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"e6cada76a2ae74af4ac383e00c2be033dfa08722","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-04-27 11:15:26.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/69/839469/2/check/DellEMC_VNX_iSCSI/2c4ded4/DellEMC_VNX_iSCSI/320 : SUCCESS in 2h 05m 15s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/69/839469/2/check/DellEMC_VNX_FC/fd212f7/DellEMC_VNX_FC/336 : FAILURE in 11m 00s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"000b9a4a8744e8a6e089149958cb531f7926179a","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-04-27 11:15:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/839469-2/HPMSA-iSCSI/ : SUCCESS in 51:58.87\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/839469-2/HPMSA-FC/ : SUCCESS in 44:05.77","accounts_in_message":[],"_revision_number":2},{"id":"d505c1899a4e1433ba1ddcc1eff146b807d0fea9","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-04-27 11:53:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/839469-2/PVME-iSCSI/ : SUCCESS in 1:07:25\n- PVME-FC https://os-logs.seagate.com/dell-ci/839469-2/PVME-FC/ : SUCCESS in 1:04:16","accounts_in_message":[],"_revision_number":2},{"id":"b7a5c5e7ef046f007e60250330b9966c897c4df1","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-04-27 11:56:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/839469-2/Seagate-iSCSI/ : SUCCESS in 1:10:01\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/839469-2/Seagate-FC/ : SUCCESS in 1:05:48","accounts_in_message":[],"_revision_number":2},{"id":"93df392022679f94790efb75492e512b3a2d3e57","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-27 12:42:44.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 55m 28s","accounts_in_message":[],"_revision_number":2},{"id":"9f868f7c0d209905d5c6b64c8aee5a599826be20","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2022-04-27 14:20:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b48690edb08ef5e06245d0f63f18835c614a655d","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2022-04-28 01:07:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b97ce03d77eaf4857c5414f5b383e3170c07957c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"date":"2022-04-28 01:10:57.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6f644c4b7c3a92c7a6b269fe854ae19ddeb6266c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-28 01:11:08.000000000","message":"Patch Set 3:\n\nBuild Started","accounts_in_message":[],"_revision_number":3},{"id":"843ff3cdc36e26083471b67877cdfcb21090ab60","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-04-28 01:27:57.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1651108267493 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"b25a773e7951cd44f8edaf7d000dbdc66363d3ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 01:39:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d78d4f8cf164838bf7aac009be434f1 : SUCCESS in 23m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c26d4bc4527a4be891d33ebe2187683d : SUCCESS in 26m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"77ec6b39da3d4c4eedd68b61606cf12294b6f428","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-04-28 02:30:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/839469 : SUCCESS in 1h 17m 39s","accounts_in_message":[],"_revision_number":3},{"id":"65023200cbd7dbc33a670e3312410075bdf8d863","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-28 02:48:01.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/69/839469/3/check/iscsi-driver-alletra-6k/8694b15/8694b15.html : FAILURE in 1h 35m 12s","accounts_in_message":[],"_revision_number":3},{"id":"101de0e270731119c7aed6cea14e19b6b1d5e218","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-04-28 02:48:51.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/69/839469/3/check/iscsi-driver-nimble/ac4ae14/ac4ae14.html : FAILURE in 1h 36m 53s","accounts_in_message":[],"_revision_number":3},{"id":"aa77e7fa8fb81213190fd1d0831fb236fd57d046","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-28 03:44:48.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/69/839469/3/check/tempest-dsvm-full-focal-py3/20d5495/ : SUCCESS in 1h 56m 33s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/9f6be4454f4f4cdcaab4336323b486b1 : DISK_FULL in 2h 07m 56s","accounts_in_message":[],"_revision_number":3},{"id":"7d1d0cbedd3a42f74a89e7f7fb834dc9be388031","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-04-28 04:15:34.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/69/839469/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b7da6a3/ : TIMED_OUT in 3h 02m 52s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/69/839469/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3baa6b4/ : FAILURE in 2h 38m 21s","accounts_in_message":[],"_revision_number":3},{"id":"5da54ec26e4611396d6ada4b5cad05894c907cec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 04:25:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4fa5d4bd5ed94056a422aef182bee00a : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a762f7f0d602400fa0cf751ed755ad6b : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83356f267cc54897adb93df8f9cc5e18 : SUCCESS in 10m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5e3e289222244b1824ed08592eb9647 : SUCCESS in 15m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d112b2eb7fe0481898d0083eee6ab794 : RETRY_LIMIT in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb9e5aff1f784006bc79d850027d9263 : SUCCESS in 10m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/0c943cae9bda4960a44830f926a8a4fe : SUCCESS in 46m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a453699c163d4772a4c09d00247b8caa : SUCCESS in 1h 06m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a974c8ad90784da2b24b5c496b8ac1b2 : SUCCESS in 41m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cf0553a864c84332bd52394c988d61c2 : SUCCESS in 12m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5278dfc032194814b66c43cf01d84182 : SUCCESS in 5m 06s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3c0d9723b0c049ac808322178af3607d : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f284616abbea49c3bf7d5b310d2d43e7 : SUCCESS in 6m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8312e5e570ee4483b7078f9364c8e448 : SUCCESS in 7m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/67d8aaab52cb442abe94bea7e24bfbab : SUCCESS in 35m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/606f56067f9c47779eb224ec031d20dc : SUCCESS in 5m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b23bc2c9b7344b77abcd56dc2f5d8e95 : SUCCESS in 1h 13m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/264e591e5b9f46f4ad6f4700ff2aba35 : TIMED_OUT in 3h 06m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6da0a0483bb5452ab941714e75ba8fda : SUCCESS in 1h 49m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/86e0f88551734094bb9d5d04e58ebb3f : SUCCESS in 49m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/955c1b02eccb4f35a72bf92f945ada6e : SUCCESS in 1h 26m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d81d8240d32b47f8bd465976f47097d9 : FAILURE in 33m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/53f40f8fb0ec46399d7e1c899fb844b2 : SUCCESS in 1h 28m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0bba79d0b190488380dd81a16ab34247 : SUCCESS in 2h 04m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35d2915dbfe84ec4b066e2e04282d246 : SUCCESS in 36m 34s","accounts_in_message":[],"_revision_number":3},{"id":"96ddc14340ccb7b6cd6d36c9028486cdadd8c6d8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-04-28 04:26:11.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/839469/3/upstream-check/cinder-cDOT-iSCSI/b716570/ : NOT_REGISTERED\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/839469/3/upstream-check/cinder-cDOT-NFS/7460aff/ : NOT_REGISTERED\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/839469/3/upstream-check/cinder-cDOT-FCP/17aee08/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"d9ae411240ca9b815a3ba852ff109a3eb06e775b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-04-28 04:26:29.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/839469/3/upstream-solidfire-check/cinder-solidfire-iSCSI/d64ba2f/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"d8f4ed0030e43c32c2c503be78ba78e487690e63","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-04-28 04:43:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-69-839469-3/1651119911 : SUCCESS in 18m 28s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":3},{"id":"16328fc86b3ee6156b3e57c02ecc1f1685f2ffb8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-04-28 04:59:49.000000000","message":"Patch Set 3:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/28_04_2022_04_25_19\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"9103ec3a3cc8877d6095194bab321101da5d2d7c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-04-28 05:34:08.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/69/839469/3/check/DellEMC_PowerFlex/0e9f351/DellEMC_PowerFlex/667 : SUCCESS in 1h 08m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"181bf0b6b5ca74249cff93312299e927f4545d94","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-04-28 05:45:41.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"4696e311087000059722c4d79d0c2130f4405ce5","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-04-28 05:48:51.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/69/839469/3/check/DellEMC_PowerStore_iSCSI/f6f25cc/DellEMC_PowerStore_iSCSI/462 : SUCCESS in 1h 23m 20s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/69/839469/3/check/DellEMC_PowerStore_FC/0be73d9/DellEMC_PowerStore_FC/477 : SUCCESS in 1h 06m 49s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"b898edcb1a9be77d97dc0d210b0171785fce1b18","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-04-28 06:09:22.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/69/839469/3/check/DellEMC_SC_FC/4c807c6/DellEMC_SC_FC/539 : SUCCESS in 57m 16s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/69/839469/3/check/DellEMC_SC_iSCSI/d4a1cdd/DellEMC_SC_iSCSI/560 : SUCCESS in 1h 43m 55s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"4b97a4c120aacf90e77cca91614766385715067a","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-04-28 06:32:30.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/839469-3/HPMSA-iSCSI/ : SUCCESS in 51:44.43\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/839469-3/HPMSA-FC/ : SUCCESS in 44:43.06","accounts_in_message":[],"_revision_number":3},{"id":"069ca20bbef30b22f3801d9c81635972f2f4f8f7","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-04-28 08:43:05.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/69/839469/3/check/DellEMC_VNX_iSCSI/6eb8fae/DellEMC_VNX_iSCSI/328 : SUCCESS in 2h 08m 33s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/69/839469/3/check/DellEMC_VNX_FC/2d16bb7/DellEMC_VNX_FC/344 : FAILURE in 12m 02s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"19d7d5e0c1077cc5d7ed8e5689181d06909c0966","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-04-28 10:03:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/839469-3/Seagate-iSCSI/ : SUCCESS in 1:08:10\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/839469-3/Seagate-FC/ : SUCCESS in 1:06:28","accounts_in_message":[],"_revision_number":3},{"id":"be4996212bb6200e75874472423a3274d67d3467","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-04-28 10:07:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/839469-3/PVME-iSCSI/ : SUCCESS in 1:08:54\n- PVME-FC https://os-logs.seagate.com/dell-ci/839469-3/PVME-FC/ : SUCCESS in 1:04:58","accounts_in_message":[],"_revision_number":3},{"id":"8b490ac6916d4f17c4b91046f8b5a0115c6c356c","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2022-04-28 11:23:42.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"353ca7aa29e9ae4e245501ab00a9499c646e4ea5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-29 09:10:55.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/69/839469/3/check/DellEMC_Unity_FC/6a51149/DellEMC_Unity_FC/702 : ABORTED in 3h 00m 00s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/69/839469/3/check/DellEMC_Unity_iSCSI/0fb1495/DellEMC_Unity_iSCSI/600 : SUCCESS in 1h 40m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"6fa2665ab99973c45055df674ecaac8eee6f96c4","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-04-29 11:15:15.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/69/839469/3/20220429175219/NEC-V-FC : SUCCESS in 1h 48m 20s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/69/839469/3/20220429175219/NEC-V-ISCSI : SUCCESS in 2h 22m 53s","accounts_in_message":[],"_revision_number":3},{"id":"8194be5323b3e80aeae239479c1940a6caf0cb14","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-04-29 14:28:53.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ad0260365b12e0aecb96faba7d40805accbfac8","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-04-29 14:28:57.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a2db5223f98ec140f22dede41accb9aecc673bc7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-04-29 14:37:09.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d3858784d823ea8a34b123b678f2923c24e03d12","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-04-29 15:30:45.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/69/839469/3/check/DellEMC_PowerMAX_FC/5396542/DellEMC_PowerMAX_FC/496 : SUCCESS in 4h 20m 35s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/69/839469/3/check/DellEMC_PowerMAX_iSCSI/b685a07/DellEMC_PowerMAX_iSCSI/501 : FAILURE in 4h 49m 45s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"cdf627fd2bbae24b0e9826585a3d666e30842e60","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-04-29 18:24:10.000000000","message":"Patch Set 3: Workflow+1 -Review-Priority\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1190968d61df8147038304051ebb5587c4ec8b3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 18:25:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9908e4ed0b4054e4adc1803552a40841c323fe97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 19:50:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fb164ff2351e4805a47a3999f6656f99 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8344d40ddcbb424a942d849eb75af850 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fc15fbd4d624a1b8e1a4194483fcf56 : SUCCESS in 12m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddef46e56d454c1d90378d040e14a9d5 : SUCCESS in 12m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f30e0b3abda45469aaa0d02096695db : SUCCESS in 14m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/701349c1fc2945d8a418d00655860fe9 : SUCCESS in 51m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/903350c20bae4879be753d1fc65c0a4c : SUCCESS in 1h 14m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f766e2b7534a49c4a14b1df6bab5f31f : SUCCESS in 48m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f3a6aa96b7bc4c90b49036aaedd07975 : SUCCESS in 1h 05m 05s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7a17cf52f36a4d57a29f3bc651ed4121 : SUCCESS in 1h 20m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36c53d812b67492d8c1f0ffa482e8a8a : SUCCESS in 1h 05m 00s","accounts_in_message":[],"_revision_number":3},{"id":"093983bbd96679316f627957a2a719db2466def7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 19:50:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ea953003141787cbaa7f8c081639ffb7bec680e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 19:53:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a7243370348494682867dfec2a29060 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3},{"id":"1689c28c6381929c0e8ab2d3b189be2d381efedd","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-02 05:20:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/839469-2/Lenovo-iSCSI/ : SUCCESS in 3:40:35\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/839469-2/Lenovo-FC/ : FAILURE in 4:29:08","accounts_in_message":[],"_revision_number":2},{"id":"f2c74958e626511ddbb3bef55a6a6d462e6ea4e2","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-04 10:26:24.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/839469-3/Lenovo-iSCSI/ : SUCCESS in 1:05:51\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/839469-3/Lenovo-FC/ : SUCCESS in 1:01:24","accounts_in_message":[],"_revision_number":3},{"id":"41a35d2172c3b3087aa527392cd12edec94aa86b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-04 16:53:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1fb4a09d43684c1faf0809d300149803 : SUCCESS in 1h 27m 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"42c6450527b2476d325e3d18382efada2c8184c6","revisions":{"c6b89c4272b02e0ebec39371253dab4c4bbd5b6d":{"kind":"REWORK","_number":1,"created":"2022-04-27 06:14:46.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/69/839469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/839469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/839469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/839469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/839469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/839469/1"}}},"commit":{"parents":[{"commit":"3e068b5ce0990a576b55fea952319884c5c0ed6d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e068b5ce0990a576b55fea952319884c5c0ed6d"}]}],"author":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2022-04-27 06:13:11.000000000","tz":480},"committer":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2022-04-27 06:13:11.000000000","tz":480},"subject":"Move centos stream testing to centos-9-stream","message":"Move centos stream testing to centos-9-stream\n\nIn Zed cycle testing runtime, we are targetting the centos 9 stream\n- https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nWith dropping the python 3.6 support, project started adding python 3.8\nas minimum, example nova:\n- https://github.com/openstack/nova/blob/56b5aed08c6a3ed81b78dc216f0165ebfe3c3350/setup.cfg#L13\n\nChange-Id: I1e0d0a43cd166e6cfd4be7ca0f426e9fc3625a28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6b89c4272b02e0ebec39371253dab4c4bbd5b6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6b89c4272b02e0ebec39371253dab4c4bbd5b6d"}]},"branch":"refs/heads/master"},"e6a786303f3309ba50c30f903e0566db833aa96b":{"kind":"REWORK","_number":2,"created":"2022-04-27 06:53:14.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/69/839469/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/839469/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/839469/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/839469/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/839469/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/839469/2"}}},"commit":{"parents":[{"commit":"3e068b5ce0990a576b55fea952319884c5c0ed6d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e068b5ce0990a576b55fea952319884c5c0ed6d"}]}],"author":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2022-04-27 06:13:11.000000000","tz":480},"committer":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2022-04-27 06:53:04.000000000","tz":480},"subject":"Move centos stream testing to centos-9-stream","message":"Move centos stream testing to centos-9-stream\n\nIn Zed cycle testing runtime, we are targetting the centos 9 stream\n- https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nWith dropping the python 3.6 support, project started adding python 3.8\nas minimum, example nova:\n- https://github.com/openstack/nova/blob/56b5aed08c6a3ed81b78dc216f0165ebfe3c3350/setup.cfg#L13\n\nChange-Id: I1e0d0a43cd166e6cfd4be7ca0f426e9fc3625a28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e6a786303f3309ba50c30f903e0566db833aa96b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e6a786303f3309ba50c30f903e0566db833aa96b"}]},"branch":"refs/heads/master"},"42c6450527b2476d325e3d18382efada2c8184c6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-04-28 01:10:57.000000000","uploader":{"_account_id":8846,"name":"Hao Wang","display_name":"Hao Wang","email":"sxmatch1986@gmail.com","username":"sxmatch"},"ref":"refs/changes/69/839469/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/839469/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/839469/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/839469/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/839469/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/839469/3"}}},"commit":{"parents":[{"commit":"3e068b5ce0990a576b55fea952319884c5c0ed6d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e068b5ce0990a576b55fea952319884c5c0ed6d"}]}],"author":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2022-04-27 06:13:11.000000000","tz":480},"committer":{"name":"wanghao","email":"sxmatch1986@gmail.com","date":"2022-04-28 01:09:28.000000000","tz":480},"subject":"Drop python3.6 support in testing runtime","message":"Drop python3.6 support in testing runtime\n\nIn Zed cycle testing runtime, we are targetting to drop the\npython 3.6 support, project started adding python 3.8 as minimum,\nexample nova:\n- https://github.com/openstack/nova/blob/56b5aed08c6a3ed81b78dc216f0165ebfe3c3350/setup.cfg#L13\n\nChange-Id: I1e0d0a43cd166e6cfd4be7ca0f426e9fc3625a28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42c6450527b2476d325e3d18382efada2c8184c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42c6450527b2476d325e3d18382efada2c8184c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
