)]}'
{"id":"openstack%2Fcinder~840511","triplet_id":"openstack%2Fcinder~master~I78915c332169b26ffb2b97310efedec65bc25e4d","project":"openstack/cinder","branch":"master","topic":"bug/1971603","attention_set":{},"removed_from_attention_set":{"34807":{"account":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"last_update":"2022-05-09 01:23:23.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-05-06 08:15:58.000000000","reason":"Rajat Dhasmana replied on the change"}},"hashtags":[],"change_id":"I78915c332169b26ffb2b97310efedec65bc25e4d","subject":"Update Volume Delete rejection message","status":"MERGED","created":"2022-05-04 16:23:06.000000000","updated":"2022-05-09 03:46:18.000000000","submitted":"2022-05-09 01:23:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"840511-bug/1971603","meta_rev_id":"187c0528b8bafe7fc452f77c44abb2f7c5cb76bd","_number":840511,"virtual_id_number":840511,"owner":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-09 01:23:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-06 08:15:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":2,"date":"2022-05-06 09:16:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":1,"date":"2022-05-06 09:16:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}],"REVIEWER":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-04 16:23:14.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-05-04 16:23:14.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-04 16:24:30.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-05-04 17:52:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-04 18:39:10.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-05-04 19:29:50.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-04 19:51:12.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-04 20:10:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-04 20:22:33.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-05-04 20:27:43.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-05-04 20:30:07.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-04 20:43:27.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-05-04 22:24:00.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-04 23:08:24.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-05-05 01:47:09.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-05-05 04:10:07.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-05-05 05:55:36.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-05 07:35:27.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-05-05 07:43:52.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-05-05 07:51:07.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-05-05 18:34:24.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-05-06 00:49:04.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-05-06 09:16:53.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2022-05-06 10:32:27.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-05-06 13:48:39.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2022-05-06 14:24:23.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-06 15:22:53.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2022-05-07 11:43:38.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-05-07 16:51:47.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2022-05-08 01:39:00.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"96e3f291d401afff2bb1fc6aa56259330b6e5c65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"date":"2022-05-04 16:23:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df265483fa29ec084f2cd49ec753bb713c027ad9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-04 16:23:14.000000000","message":"Patch Set 1:\n\nBuild Started","accounts_in_message":[],"_revision_number":1},{"id":"dacfa84e7aca6e21899f786c62e02b80dbd558df","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-04 16:23:14.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15939/","accounts_in_message":[],"_revision_number":1},{"id":"e7ab60d4002c559f9e9e0fda40e7abf822bd18fe","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-04 16:24:30.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"0b0c7e1e24adbce07b35abd0b6807b7e1dbd80ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"date":"2022-05-04 16:24:58.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e82a0b13f08e02d37e98b5badfc6d136ca88aed2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-04 16:25:02.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15939/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d158490916979513e5256a20b59aa2605c8794ee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-04 16:25:09.000000000","message":"Patch Set 2:\n\nBuild Started","accounts_in_message":[],"_revision_number":2},{"id":"1f34bdd5aa32a555d1e8f0ca446de485d84c908e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-04 16:26:34.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"a86a723e8b5e4592695fbebc5f2fb48024b92cc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"date":"2022-05-04 16:32:03.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"90050983be035b0ce71abf37bbeb0aa21cd71604","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-04 16:32:14.000000000","message":"Patch Set 3:\n\nBuild Started","accounts_in_message":[],"_revision_number":3},{"id":"e977fe964efac809ef90e5814780e2b8e8a3d5e4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-04 16:33:13.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"b235b6612953b476c4c047d8e08c9f8dc5648c5c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-04 16:41:35.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1651681393393 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a02ec22f3d50f74226297134db7502c4e3de9735","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-04 16:59:07.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1651681508393 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"1547e34772d774cc8ff4719a766a419f54dc5dda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"date":"2022-05-04 17:09:11.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"5261676ecd9d3a8a4c11d05abb0583fc666aef6e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-04 17:09:19.000000000","message":"Patch Set 4:\n\nBuild Started","accounts_in_message":[],"_revision_number":4},{"id":"475e519b3e2160f6ddb4d123d7e2bc453bd7fd99","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-04 17:09:46.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"002f51543fbe0a7d245ddf7c42f86b71478c0ba8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-04 17:16:50.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1651681933421 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"fcb5f23df1ab496eb79bb7ff2e56422baa66f6a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 17:52:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe7573ee20bb43ea921a370e8054b394 : SUCCESS in 22m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be6dc5aea1c849b496357d17f9b81d07 : SUCCESS in 24m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0c0e3e4389398030d0b33e7255e32da6e0dfa878","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-04 18:39:10.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/840511/4/iscsi/ : FAILURE in 1h 29m 26s\n- cinder-smb http://cloudbase-ci.com/cinder/840511/4/smb/ : SUCCESS in 1h 16m 23s","accounts_in_message":[],"_revision_number":4},{"id":"e1bb37cccd68a9381588adaf4ebbdb8b64ab6abe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-04 18:46:19.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1651684158559 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"0df93cbfaf736ab224e8496d0aca667bd94d902c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-04 19:29:50.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/840511 : SUCCESS in 1h 18m 32s","accounts_in_message":[],"_revision_number":4},{"id":"d2c2522ffb8e5127492a8c77ecd912460fa9e592","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-04 19:51:12.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-11-840511-1/ns5_nfs : SUCCESS in 3h 27m 34s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-11-840511-1/ns5_iscsi : SUCCESS in 3h 26m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"b9dfb2c11e38d88fb4e4a0d01a6ee86bba7e551d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 20:10:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/415dd2c52b064a5987ebcf717b914660 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab8f6f14f48242b098301b22ac062fed : SUCCESS in 10m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9688b5677c5448858a41e78f7e9cd460 : SUCCESS in 12m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35938ad8f4034ba48eab409917fc94a7 : RETRY_LIMIT in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3d2ef9aab0c4ff2b36c8112b28e8025 : SUCCESS in 11m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/123ea4505cc947ccb41b0735936d8c08 : SUCCESS in 1h 02m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d233b16dc99a47da963d3a229c65ebd8 : SUCCESS in 1h 31m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c23853187714e64a72c3494d1076345 : SUCCESS in 58m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/98dfb62461ee42439fb3b16ca6ca5522 : SUCCESS in 15m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3748ab9cd78f446baa6217e6234007c2 : SUCCESS in 5m 06s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b4ae7b9d02af48b39da93947346ecc01 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/136835f2c9094931bf280151f017d3c3 : SUCCESS in 6m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7d90e31d727b49d6aa38395bffe20b3d : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b105743872884311bea744ec03e49b99 : SUCCESS in 47m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8354a7f9c1224beeb3f83b3b4c1522cb : FAILURE in 6m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/945b565bfeef4321b14873b8a5be8bb4 : SUCCESS in 1h 21m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6e6ccb54bc36410580085ad2904627a4 : FAILURE in 1h 40m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/45d00793eb564c2d925d17bb7530ac84 : SUCCESS in 2h 01m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/12946ba3af13480fa7c0eacd6c208ae6 : SUCCESS in 1h 14m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/94a3466fbce24169bf63d172aa0a8c2d : SUCCESS in 1h 34m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/107ee734c0dd4c45aa43a4e847a46ee6 : SUCCESS in 55m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/23fccd2da330448d84bfdff905e1cd3c : SUCCESS in 1h 12m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c65ab9bdf7b640cb88958786519b0a34 : SUCCESS in 2h 53m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/217785429b114cada8ec83ebba7db09a : SUCCESS in 58m 54s","accounts_in_message":[],"_revision_number":4},{"id":"c090766863cc4f01f48db7b3663240314d15aa55","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-04 20:16:01.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-11-840511-2/ns5_nfs : SUCCESS in 3h 24m 59s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-11-840511-2/ns5_iscsi : SUCCESS in 3h 29m 49s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"d91b9d4828cda39381ba37273f03fb4af5e1f4c7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-04 20:22:33.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/11/840511/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f6b27ce/ : TIMED_OUT in 3h 02m 56s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/11/840511/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/068fa3d/ : TIMED_OUT in 3h 03m 14s","accounts_in_message":[],"_revision_number":4},{"id":"35dde871390915d9920a2834a1a8c801e9d6996b","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-05-04 20:27:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-11-840511-4/1651695012 : SUCCESS in 17m 29s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"ffb2a439238ef9a36a3aba31db5070e01c3eba81","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-04 20:30:07.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/11/840511/4/check/tempest-dsvm-full-focal-py3/f9b3b9f/ : SUCCESS in 2h 07m 32s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/11/840511/4/check/devstack-plugin-ceph-dsvm-focal-py3/5560006/ : SUCCESS in 2h 34m 11s","accounts_in_message":[],"_revision_number":4},{"id":"9b427dd4b3d6ca46cc2b8083b5e1091c610992d6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-04 20:38:43.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-11-840511-3/ns5_nfs : SUCCESS in 3h 25m 51s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-11-840511-3/ns5_iscsi : SUCCESS in 3h 29m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"c51610221006c83276c5845440952e93ec72336b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-04 20:43:27.000000000","message":"Patch Set 4:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/04_05_2022_20_16_02\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"a8288c43cd371892e6f018d156da17562bf40cf9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-04 22:24:00.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/11/840511/4/check/DellEMC_PowerFlex/72ba7d9/DellEMC_PowerFlex/695 : SUCCESS in 1h 12m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"c2e824f7b81be42c6e3affb02857396f0fba3e55","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-04 22:31:17.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-11-840511-4/ns5_nfs : SUCCESS in 3h 27m 39s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-11-840511-4/ns5_iscsi : SUCCESS in 3h 29m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"11188128611195fc1927ded71ea327606e54f65e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-04 23:08:24.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/11/840511/4/check/iscsi-driver-nimble/cf71f01/cf71f01.html : SUCCESS in 1h 37m 12s","accounts_in_message":[],"_revision_number":4},{"id":"7970c386724548d974d5c23206a1232be2bf8921","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-05-05 01:47:09.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/11/840511/4/check/DellEMC_SC_FC/2143269/DellEMC_SC_FC/567 : SUCCESS in 1h 03m 47s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/11/840511/4/check/DellEMC_SC_iSCSI/9b3bb7f/DellEMC_SC_iSCSI/588 : SUCCESS in 1h 46m 21s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"c5baccb2414f0233e1723a50d7ffdcf88c636145","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-05-05 04:10:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/840511-4/HPMSA-iSCSI/ : SUCCESS in 50:36.49\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/840511-4/HPMSA-FC/ : SUCCESS in 46:09.34","accounts_in_message":[],"_revision_number":4},{"id":"15d778250d9269dfd02c8be88be80ec18a2b197f","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-05 05:55:36.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/11/840511/4/check/DellEMC_VNX_iSCSI/f5c81ee/DellEMC_VNX_iSCSI/356 : FAILURE in 2h 55m 38s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/11/840511/4/check/DellEMC_VNX_FC/8d0eba5/DellEMC_VNX_FC/372 : FAILURE in 9m 12s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"d5feddf07f9be66972c6b082c79c7a47423e650b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-05 07:35:27.000000000","message":"Patch Set 4: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"10a2c50bbfc28fabae1e1fb31da0610ac51bca21","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-05-05 07:43:52.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/840511-4/Seagate-iSCSI/ : SUCCESS in 1:08:41\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/840511-4/Seagate-FC/ : SUCCESS in 1:03:57","accounts_in_message":[],"_revision_number":4},{"id":"eba7269c63567ccec3b76207189c571f049a9f6c","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-05-05 07:51:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/840511-4/PVME-iSCSI/ : SUCCESS in 1:04:30\n- PVME-FC https://os-logs.seagate.com/dell-ci/840511-4/PVME-FC/ : SUCCESS in 1:04:08","accounts_in_message":[],"_revision_number":4},{"id":"66256f3cd54c143de5fce0fb0b42c8ee168c3fc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"date":"2022-05-05 17:17:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"70491088067b00a4532f5a6f3e44c0d6063ab468","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-05 17:17:36.000000000","message":"Patch Set 5:\n\nBuild Started","accounts_in_message":[],"_revision_number":5},{"id":"a84a55d9f44333530da44b10a7d396c23c456ba0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-05 17:17:41.000000000","message":"Patch Set 5:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15955/","accounts_in_message":[],"_revision_number":5},{"id":"083ff2c75b114bbe7edef9e2c108f9bc1acd4f53","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-05 17:19:30.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15955/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"dfa373bb4082d84e738dfcb3416e274bbfe296cb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-05 17:21:34.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"5f0a471d54d511dd8f149948c8f02cbca91efc88","author":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"date":"2022-05-05 17:27:33.000000000","message":"Patch Set 5:\n\n(8 comments)","accounts_in_message":[],"_revision_number":5},{"id":"40efccbfdcd4750862575142a247b1485bef20ba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-05 17:35:14.000000000","message":"Patch Set 5:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1651771056166 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"02b72099294c24374c77dff61489454c9f941395","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 17:40:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be0759ebcfc643b7902083e8a77a21f3 : SUCCESS in 20m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4747ca122fef40ce8e32663399fc8d65 : SUCCESS in 19m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8253b000a0565047efe89c4a21c04b539ba0ab8b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-05 18:18:01.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/11/840511/5/check/iscsi-driver-nimble/203fde9/203fde9.html : FAILURE in 59m 22s","accounts_in_message":[],"_revision_number":5},{"id":"12a940c11efcd2cebbc142b2a85be2dfe198eda9","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-05 18:34:24.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/11/840511/4/20220505202629/NEC-V-FC : SUCCESS in 1h 47m 21s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/11/840511/4/20220505202629/NEC-V-ISCSI : SUCCESS in 7h 07m 52s","accounts_in_message":[],"_revision_number":4},{"id":"96fdaf37074c4cc205d1b63c6d401be57d23e3c5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-05 18:37:19.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/840511 : SUCCESS in 1h 18m 05s","accounts_in_message":[],"_revision_number":5},{"id":"5647815a170804e504ae63aa23795b1525940aa0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-05 18:37:58.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/840511/5/iscsi/ : SUCCESS in 1h 19m 58s\n- cinder-smb http://cloudbase-ci.com/cinder/840511/5/smb/ : SUCCESS in 1h 09m 34s","accounts_in_message":[],"_revision_number":5},{"id":"b78cd055b2c5ee50154bd41b221adc5053407489","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 19:59:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ba6b45a97e49ff8f91ea285031a0fa : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4749d2eeefb4d8cac47fa198895a973 : SUCCESS in 10m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be97334fee314359910a2bd4851a33b6 : SUCCESS in 11m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9644851ce214a50aab8f3d7e990d840 : RETRY_LIMIT in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/470a9edbe23540e5aff0f04a30ebcd05 : SUCCESS in 13m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/4ef2fab2d20c46db957fd57d6d1ab221 : SUCCESS in 1h 06m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9a13aef99dac459bb3e9627de5f01a38 : SUCCESS in 1h 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4dd30e4da90c4b59aa9a42afc4150fa7 : SUCCESS in 53m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bd08f9bcb6b64406bab5d8ab8d902d34 : SUCCESS in 15m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/55f6a3f63b8a4befb96ac374b2d77164 : SUCCESS in 5m 52s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b0633d35f5e44cf5ba6c4e375cb0e754 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a359c95fd25e414d897f94d27d40e9e3 : SUCCESS in 6m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7b0b19459b9e4bcca40e7733755f7675 : SUCCESS in 6m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/68444939778f4e98a0caaf50615016fc : SUCCESS in 47m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/77259a4a2b134c60a53570d89823d146 : FAILURE in 6m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f2b1e0b9b1134512b524f6a3c11dc850 : SUCCESS in 1h 11m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cb00b1d6d84d44cc9b69789e5a50a349 : FAILURE in 1h 42m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d2d8df505fd44b03b11dddcc12ab3eac : SUCCESS in 1h 53m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/48e6f2a3126b4463b60fb2bf4d138dcb : SUCCESS in 1h 09m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2e35f32a919940f1aa2701974130378f : SUCCESS in 1h 34m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d52fd69361e64287866cb3bd6c3964f1 : SUCCESS in 47m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3b4437933315411190d5802464da9006 : SUCCESS in 1h 23m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2eadeaab76fc44778f9fe50ebecbe450 : SUCCESS in 2h 34m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69aea944c2584109b4ff94321ea2fa70 : SUCCESS in 53m 37s","accounts_in_message":[],"_revision_number":5},{"id":"544f149f16abd7c7ae4d729ac8d8065b2a9d0cd5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-05 20:16:37.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/11/840511/5/check/tempest-dsvm-full-focal-py3/57982ef/ : SUCCESS in 2h 57m 16s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/11/840511/5/check/devstack-plugin-ceph-dsvm-focal-py3/2e0044d/ : SUCCESS in 2h 06m 46s","accounts_in_message":[],"_revision_number":5},{"id":"b4b7d8e8ee8523f580c233712f1b51d18d33d0b8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-05 20:23:07.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/11/840511/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c9e8cf7/ : TIMED_OUT in 3h 03m 45s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/11/840511/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/30ea885/ : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":5},{"id":"c3a55f482498067eddceb6d2d1b5e3bc262c2663","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-05 20:32:37.000000000","message":"Patch Set 5:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/05_05_2022_19_59_29\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"3766b28f7ade8e42d662f3534312caf0f6ec3873","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-05 20:47:24.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-11-840511-5/ns5_nfs : SUCCESS in 3h 27m 29s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-11-840511-5/ns5_iscsi : SUCCESS in 3h 29m 23s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"bf5b54a55fe118ea9af4e878a59b85929b606813","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-05 21:49:37.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_PowerFlex/6d5add8/DellEMC_PowerFlex/701 : SUCCESS in 1h 19m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"40de6816705e25bb91ea377246689bf06bf7828e","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-05-06 00:10:04.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_SC_FC/1507f6d/DellEMC_SC_FC/573 : FAILURE in 21m 25s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_SC_iSCSI/276fc32/DellEMC_SC_iSCSI/594 : SUCCESS in 1h 45m 59s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"962976aff6d2d586993325aa083d24cf362b808e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-06 00:49:04.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_Unity_FC/e077544/DellEMC_Unity_FC/739 : FAILURE in 30m 23s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_Unity_iSCSI/e12f400/DellEMC_Unity_iSCSI/657 : SUCCESS in 1h 43m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"1f55bb24bca1df5d8983a8ed3bdd64a652ae380c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-06 01:27:00.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_VNX_iSCSI/577ef6b/DellEMC_VNX_iSCSI/361 : SUCCESS in 2h 16m 15s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_VNX_FC/cddbd63/DellEMC_VNX_FC/378 : FAILURE in 13m 24s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"ec15c95c02a9264e8173002abef50852ea1aea0d","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-05-06 01:40:13.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/840511-5/HPMSA-iSCSI/ : SUCCESS in 52:03.38\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/840511-5/HPMSA-FC/ : SUCCESS in 44:09.41","accounts_in_message":[],"_revision_number":5},{"id":"72b52288ae0ad6e646052040404e81dac042bf21","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-05-06 03:52:27.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/840511-5/PVME-iSCSI/ : SUCCESS in 1:05:43\n- PVME-FC https://os-logs.seagate.com/dell-ci/840511-5/PVME-FC/ : SUCCESS in 1:04:55","accounts_in_message":[],"_revision_number":5},{"id":"14de2df53e55a5682754550fad05e676662f3ba1","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-05-06 04:16:51.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/840511-5/Seagate-iSCSI/ : SUCCESS in 1:08:08\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/840511-5/Seagate-FC/ : SUCCESS in 1:06:45","accounts_in_message":[],"_revision_number":5},{"id":"ab6b97d5b1b5c890a1c199edcff4e8f0a3e0ea96","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-06 08:15:58.000000000","message":"Patch Set 5: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"6cfef5da76fc94a1df900eb9519ba59bb0310627","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-05-06 09:16:53.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"2690d23257907028bdcd34ce41ab5988f80cbc17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 09:18:07.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"cba22cdb7abcc29f6d76f62a821e228be4c6f6c4","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-05-06 10:32:27.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_PowerStore_iSCSI/9bbf806/DellEMC_PowerStore_iSCSI/486 : SUCCESS in 1h 19m 45s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_PowerStore_FC/c729eca/DellEMC_PowerStore_FC/511 : SUCCESS in 1h 11m 08s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"04e43b4159274cb89b50f084fcd3c599765da453","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 12:39:44.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e782ccba214b4f8e844c44cfb866263c : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ee1fc9d70ee4d3d8c7ea77ef3817b55 : SUCCESS in 11m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39c8cba2b733465388a06859b47e4136 : SUCCESS in 11m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2510fece8edf4cb1a57e95869b17679c : SUCCESS in 12m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/3148b67d6aab40e7b2447cf5c2592cf7 : SUCCESS in 48m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bb06bd9a52d340508aa4f17575f9c45f : SUCCESS in 1h 13m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2e5cc87b08e48b194d0e15e24652ae4 : SUCCESS in 48m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d184bbe851a3482ab3b3aa5ea81c68ab : SUCCESS in 1h 04m 36s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/051fb0e1da374e2281439b879b41a1f9 : FAILURE in 1h 13m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d38e04c4e22a482e9009612d9fc58e5c : SUCCESS in 44m 49s","accounts_in_message":[],"_revision_number":5},{"id":"a16eacc450c2648ce87623d5b6226940eec889ed","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-06 13:48:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ddb6cf02f96595fd22aaccc4aed1511e86ac5ff9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-06 13:53:07.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"c1517b2ba69b66ad3a2c39d1a1ab6af684bab648","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-06 14:24:23.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2a3e783c15bf467abd9011bdf10d88bf : SUCCESS in 1h 30m 17s","accounts_in_message":[],"_revision_number":5},{"id":"fe096131919e9cdeec7d85f0d667275d4508b0da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 14:35:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/daf3b3d3555c461e92e8543ec28ee42d : SUCCESS in 20m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/07d453d861764577a7fd563201942cd2 : SUCCESS in 20m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f71d8a0c5be833bc4e09a32c8a0d9b6b5a002c2d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-06 15:12:31.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/840511/5/iscsi/ : SUCCESS in 1h 12m 30s\n- cinder-smb http://cloudbase-ci.com/cinder/840511/5/smb/ : FAILURE in 1h 23m 25s","accounts_in_message":[],"_revision_number":5},{"id":"3565994f743fb48482b5c9b7d3acf00960073df8","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-05-06 15:22:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/97dc2a8/index.html : SUCCESS in 1h 27m 31s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/194604b/index.html : SUCCESS in 1h 33m 16s","accounts_in_message":[],"_revision_number":5},{"id":"4aae60e83c750c6ada4d35faff73ed8a0944d51a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 16:23:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05668333b9c44a66a3a2d7c743dfabf0 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6766d0115e74a62998311c62f809d73 : SUCCESS in 10m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3a13ad3fb58401a8d1a8cb0268078be : SUCCESS in 11m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96ba9cbd624f474790c1802af6b5bf59 : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f676284abed4227896e105527d85e30 : SUCCESS in 11m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/46c4b853e8234436bad70f0e12151c78 : SUCCESS in 59m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1cc6a7e3829e4477b2524881759f9bec : SUCCESS in 1h 19m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74c9d794460b4ee0a5a8b1c7191f900a : SUCCESS in 57m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/47d6a07563094610af17ba24c647ccac : SUCCESS in 16m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/18a32c624c6d45f48ca1447418a7a201 : SUCCESS in 5m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/152c9ec7375240c09248a5c7517b9d99 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e296ac6d1760455da5a1ba0b38cb275b : SUCCESS in 6m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3b9c9b5510ca47d78bdc18fd1d4b7eff : SUCCESS in 6m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eaa42e3715f84897a5adaa8de0fc50ee : SUCCESS in 45m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a9a0d0fdf456474293508adb155b1a68 : FAILURE in 6m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3eff91c42da34e3aaa9c52310a43e51c : SUCCESS in 1h 17m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d88ae8f83f4641fabd6bc3f78f27a3d6 : FAILURE in 1h 34m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a699ea62a0d94aabba07a1bdbfda0749 : SUCCESS in 1h 41m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2efebcadd2534659a661c3084a68dbd6 : FAILURE in 39m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3930386f6bf5413fae19cf5b21fe5df4 : SUCCESS in 1h 24m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a51b15bd3fb447cfa5dc1f3e3f841cf6 : SUCCESS in 33m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/99759c2eff1243e08601f959bc41d15a : SUCCESS in 1h 09m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e75c0dbbb8714541a58922585c5d29a0 : SUCCESS in 2h 28m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cc2aa7d3806497794c2d1be7a0aa9c8 : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":5},{"id":"d1af49d64efbce74cbd5cc9e137d37a13d4ef247","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-06 17:03:26.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/11/840511/5/check/tempest-dsvm-full-focal-py3/356e3e5/ : SUCCESS in 1h 55m 08s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/11/840511/5/check/devstack-plugin-ceph-dsvm-focal-py3/30bf858/ : TIMED_OUT in 3h 10m 21s","accounts_in_message":[],"_revision_number":5},{"id":"70be1a00db5fb9c8e8a5f267da7a205baa9dcd12","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-06 21:46:44.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/11/840511/5/20220507043433/NEC-V-FC : SUCCESS in 1h 55m 30s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/11/840511/5/20220507043433/NEC-V-ISCSI : SUCCESS in 2h 12m 08s","accounts_in_message":[],"_revision_number":5},{"id":"c80dac4a5ed9b5f56b2c67df981442dfc6fa76d9","author":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"date":"2022-05-07 08:33:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ee04a5b248b5eb3d1fb8fe4abc1c64bed4bec3b5","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-05-07 10:01:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-iscsi/1778f53/index.html : SUCCESS in 1h 25m 24s\n- storwize-tempest-dsvm-full-fc http://s3.eu-geo.objectstorage.softlayer.net/ibm-storage-ci-log/storwize-tempest-dsvm-full-fc/0482092/index.html : SUCCESS in 1h 26m 30s","accounts_in_message":[],"_revision_number":5},{"id":"d47863575ccdc4011790572ef5d04ebd61937646","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-07 11:43:38.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/840511-4/Lenovo-iSCSI/ : SUCCESS in 1:03:46\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/840511-4/Lenovo-FC/ : SUCCESS in 58:21.80","accounts_in_message":[],"_revision_number":4},{"id":"8072818f8dbd026757d4dc9911bc9a4367db6793","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-07 16:51:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e1d624ad7aed181bc46f53f59e0873a86fb210f7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-07 16:53:53.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"4dd03bf1f59afa181c2df0982b665ffccb4a78b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 17:13:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c43904716e74badbc317c3419dc683f : SUCCESS in 18m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6da7810abe3949a58946042ce1a33e07 : SUCCESS in 19m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"abe536f978fb039872c574ac80a7ecac78e2a56b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-07 17:56:24.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/840511/5/iscsi/ : SUCCESS in 1h 04m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/840511/5/smb/ : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":5},{"id":"0efabe51997fb98cea5694403d18f8668d4e61b1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-07 18:26:06.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/13e11437024b4cc7a51fe14102346333 : SUCCESS in 1h 33m 10s","accounts_in_message":[],"_revision_number":5},{"id":"b890d19a6dc67af51b3846ec5eb63c4afa612ffc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-07 19:03:11.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/11/840511/5/check/tempest-dsvm-full-focal-py3/d4a65e4/ : SUCCESS in 1h 33m 11s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/11/840511/5/check/devstack-plugin-ceph-dsvm-focal-py3/7fce833/ : SUCCESS in 2h 09m 07s","accounts_in_message":[],"_revision_number":5},{"id":"21d3ba22ad51fee79d29aa879f9fa5bbacb4ca21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 19:14:00.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2caa9650dcee422ea2484c03a01f36ed : SUCCESS in 8m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a62756af690a46dfab0041ecc43a91c2 : SUCCESS in 10m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03b4e4a41dff49f1ba938bfac93c11e7 : SUCCESS in 11m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cce6fc8ae3734bd6a4625a1c00e727d8 : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8209bb2eef44e4584157c3ae1dd55c4 : SUCCESS in 12m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/f3aa52ed8f5048398fd360f8077e6e16 : SUCCESS in 46m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/67b90c7afcc94e09b44ec16a79fb29ce : SUCCESS in 1h 03m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecab9437d1ef4c11bcc16b7d8403b54d : SUCCESS in 38m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ca2aa4e91d18465fbcbeb61513a7d8c4 : SUCCESS in 14m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/37b50815b4194f8c93331c7e5326606e : SUCCESS in 5m 13s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4e286bc5b69d4c78902b86728abda675 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d8902bd463ff41d392b98729fc38cce9 : SUCCESS in 5m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7db89c8796b3407aa4833893a5704506 : SUCCESS in 7m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/177b5cd5bed944648a1feb6a8006e4b2 : SUCCESS in 29m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/705a97ee818941c098dd50c964be58d5 : FAILURE in 7m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/10e951a892a8482dba58568f134173cd : FAILURE in 39m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dc64c2b7dfd746c0b4cdc8784b451dff : FAILURE in 1h 17m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/34d0bdefd1a74919af8bd8754cd1cd4e : SUCCESS in 1h 37m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/418ccb786e97416e9cee1b1860de4658 : SUCCESS in 47m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f1dc87259bb54bf0bb5bd7bedb4b6751 : SUCCESS in 1h 00m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1e1b5883bd224abd8d2ca13e6b8e8263 : SUCCESS in 40m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/916a38c45ebc4a60b054b6a041d605d4 : SUCCESS in 1h 00m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/747124286f4743f095dc61af08344cd1 : SUCCESS in 2h 20m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f79bdce79a8450ab2ec904101f39b8e : SUCCESS in 45m 11s","accounts_in_message":[],"_revision_number":5},{"id":"060f3ad4e5bbcee63dd7764bf7540a8ccc690ac7","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-05-08 01:39:00.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/11/840511/5/20220508041825/NEC-FC : SUCCESS in 3h 16m 32s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/11/840511/5/20220508041825/NEC-ISCSI : SUCCESS in 6h 20m 31s","accounts_in_message":[],"_revision_number":5},{"id":"fd9501510b2e0de6fcfa1d74329338e05643ca5a","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-08 03:10:05.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/840511-5/Lenovo-iSCSI/ : SUCCESS in 1:03:17\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/840511-5/Lenovo-FC/ : SUCCESS in 59:59.32","accounts_in_message":[],"_revision_number":5},{"id":"cd0776f2d76b655c005ee0aa33907fc43ae18cf2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-05-08 21:21:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"66b6f1c3b9fa3508a928419074e4d7b9a21c6070","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-08 21:41:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e99ce4e658c944b5956cd32a3d501541 : SUCCESS in 17m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9e51475d27ce4bd3a557fc8e53a7e0d2 : SUCCESS in 18m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c8f2da9e47abab8ac0ee97f11fec41493f91c6ed","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-08 22:33:19.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/840511/5/iscsi/ : SUCCESS in 1h 11m 41s\n- cinder-smb http://cloudbase-ci.com/cinder/840511/5/smb/ : SUCCESS in 1h 01m 37s","accounts_in_message":[],"_revision_number":5},{"id":"856272aeed80dabffae862a077458191354153a1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-08 22:53:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3cd6a99e1b594f68bd2af18ffd51239a : SUCCESS in 1h 31m 23s","accounts_in_message":[],"_revision_number":5},{"id":"ec439c41821bf07742334fb77b86ff413b6f287a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-08 23:27:27.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/11/840511/5/check/tempest-dsvm-full-focal-py3/e7853cd/ : SUCCESS in 1h 33m 15s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/11/840511/5/check/devstack-plugin-ceph-dsvm-focal-py3/780b41d/ : SUCCESS in 2h 04m 14s","accounts_in_message":[],"_revision_number":5},{"id":"371c9e28aa4b95686037fefd075c0caf9da44f6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-08 23:53:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f1274ea570146f5ba76c6a82f8e1a0f : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3522cf7a9b054394a02b0d11fdd4e17f : SUCCESS in 10m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bfe2935e975460e8d99162e775e6141 : SUCCESS in 11m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a55bff1c87144309ad1bd95618616ab : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6af5f2fe4d3a4a64b3d125af77311a3c : SUCCESS in 12m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/904f71d007ff4211957a3081f37cd2ca : SUCCESS in 54m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4771bdb35eb14bc2bb7efb86201f4102 : SUCCESS in 1h 22m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dec062248e3d4259adadfc4f263445c3 : SUCCESS in 48m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/090171c90d494bdf89bb939a72ab0622 : SUCCESS in 13m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a5eaad350779499685cc383fea7148d1 : SUCCESS in 5m 26s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1e1ba44029124eb3b250563c45f2ab2e : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a382512c547d4f79b855ee47e73ef856 : SUCCESS in 6m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f87d398d69dd4388bf693b44c512db44 : SUCCESS in 7m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/df627a5dd6bf4d1d88479a876b922840 : SUCCESS in 27m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9480b993d3e24d798845fd8ea6dd9eb1 : FAILURE in 7m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fa6efa4db9144e8b805e16c96f3a787c : SUCCESS in 1h 30m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/df7f541620ba4e7681bcff11dca83f03 : FAILURE in 1h 34m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3805a950e3b3437e9766feda0be14f5b : SUCCESS in 2h 05m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e8ae46e50a4842e5998a0d9b8789af56 : SUCCESS in 1h 14m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7b0a36292ba240fcb44d1f5ceb637ed7 : SUCCESS in 1h 53m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/02b7f7b9cd904517bd464ea74a31c0ca : SUCCESS in 48m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1bf880a51bf946e1a9b61d7da903d229 : SUCCESS in 1h 30m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b228f21643694d16a6e3f1ad5319202b : SUCCESS in 2h 27m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/955b83c3deb34871ab0462910ab1770d : SUCCESS in 1h 02m 27s","accounts_in_message":[],"_revision_number":5},{"id":"734a9d8a2d39f239545a225651d6ddfedad5940e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-08 23:54:07.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3f2372aa4490c3d6ba3b6cc3bcaac22b1f00e75c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-09 00:19:39.000000000","message":"Patch Set 5:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/08_05_2022_23_53_29\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"3f35354c3e8acb275733b3b151f479e6e4731411","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-09 01:01:14.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_PowerFlex/6b139c5/DellEMC_PowerFlex/703 : SUCCESS in 1h 07m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"2f6b1080cff105376c80b5e616e83db69b154df9","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-05-09 01:08:43.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_PowerStore_iSCSI/f4e7535/DellEMC_PowerStore_iSCSI/488 : SUCCESS in 1h 15m 04s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_PowerStore_FC/81435ad/DellEMC_PowerStore_FC/513 : SUCCESS in 1h 10m 35s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"2e94eb63ccf76448be34f4b80904f0835dff9dfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 01:23:23.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca33f065b52d4ac593e87b9987708714 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60a8f3b9348140e18e3112d2f889c278 : SUCCESS in 12m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e76b0ac5a304365b5a0455afd08ead7 : SUCCESS in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/169e754a987440cf9e8829ced8b8888d : SUCCESS in 10m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/0c748dee9c62402c8938b0e05c63914b : SUCCESS in 49m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/399c83b5c2944d02bc293cce40488745 : SUCCESS in 1h 09m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb89928b9e8844cc887661aa04e58b97 : SUCCESS in 42m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bf60e917b4d54903a51a44d9836171f2 : SUCCESS in 54m 04s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c9fd4fba127142fd93d6b0eef5ae8634 : SUCCESS in 1h 23m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7586068b0ec24998a4d4e2ee90712f23 : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":5},{"id":"d52ec630086d8c56c30744441eb97eb0025611ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 01:23:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"3aaf8ac39e7ed685db7c08b63b5b6ef1dea53124","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 01:24:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95b0ee3d61314b478af732c60e77ec8c : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":5},{"id":"3b0fbd14b8958ef95e9b828407bb86963efbf708","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-05-09 01:36:46.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_SC_FC/45bd71d/DellEMC_SC_FC/575 : SUCCESS in 56m 27s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_SC_iSCSI/6d86c45/DellEMC_SC_iSCSI/596 : SUCCESS in 1h 42m 23s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"4f0cd200bd1cddc5201a5fcef72054d8598b00b2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-09 01:39:42.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_Unity_FC/2e1ba25/DellEMC_Unity_FC/742 : SUCCESS in 1h 15m 11s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_Unity_iSCSI/40dac7c/DellEMC_Unity_iSCSI/660 : SUCCESS in 1h 46m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"015401ead4bd6042ed5780703548f6c4a4624d19","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-09 01:59:53.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_VNX_iSCSI/5892886/DellEMC_VNX_iSCSI/363 : SUCCESS in 2h 06m 16s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/11/840511/5/check/DellEMC_VNX_FC/8a57025/DellEMC_VNX_FC/380 : FAILURE in 10m 25s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"5fe0d877a4a320573b220105b0b988cc2cf7e366","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-05-09 02:03:38.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/840511-5/HPMSA-iSCSI/ : SUCCESS in 52:24.11\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/840511-5/HPMSA-FC/ : SUCCESS in 46:17.29","accounts_in_message":[],"_revision_number":5},{"id":"a6145fbf34caa80ea51bb060de8121a8113b2a2b","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-05-09 02:41:15.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/840511-5/Seagate-iSCSI/ : SUCCESS in 1:08:06\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/840511-5/Seagate-FC/ : SUCCESS in 1:08:17","accounts_in_message":[],"_revision_number":5},{"id":"6097132506a720dd9f37c98b8338e4be3f255c2e","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-05-09 02:41:22.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/840511-5/PVME-iSCSI/ : SUCCESS in 1:08:29\n- PVME-FC https://os-logs.seagate.com/dell-ci/840511-5/PVME-FC/ : SUCCESS in 1:05:07","accounts_in_message":[],"_revision_number":5},{"id":"6ecc8966d6b21cc4593036396089e3f81b4225eb","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-09 02:53:06.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/840511-5/Lenovo-iSCSI/ : SUCCESS in 1:14:05\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/840511-5/Lenovo-FC/ : SUCCESS in 1:11:05","accounts_in_message":[],"_revision_number":5},{"id":"187c0528b8bafe7fc452f77c44abb2f7c5cb76bd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-09 03:46:18.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2b75fa3021e1e6b4d635ab36a9084719c0e3d969","revisions":{"12f06f7858b44bc79b1d023c1d98cdfa193d5bcf":{"kind":"REWORK","_number":1,"created":"2022-05-04 16:23:06.000000000","uploader":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"ref":"refs/changes/11/840511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/840511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/840511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/840511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/840511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/840511/1"}}},"commit":{"parents":[{"commit":"a5405554577f16c7d498ef55844d00b1fce6aa9d","subject":"Merge \"[SVf]:Fix multiple lsvdisk calls for GMCV create volume operation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a5405554577f16c7d498ef55844d00b1fce6aa9d"}]}],"author":{"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","date":"2022-05-04 15:29:51.000000000","tz":0},"committer":{"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","date":"2022-05-04 15:29:51.000000000","tz":0},"subject":"Volume Deletion rejection message  does not include \"awaiting-transfer\" in invalid state","message":"Volume Deletion rejection message  does not include \"awaiting-transfer\" in invalid state\n\nWhen user tries to delete a volume in \"awaiting-transfer\" state ,\nthe error message received does not include the \"awaiting-transfer\"\nstate in invalid states of Volume for deletion.\nThis leads to the user believing the volume-delete request is valid\nand is not able to debug the reason for failure of the volume deletion\nfor volume present in \"awaiting-transfer\" state.\n\nCloses bug: Bug #1971603\n\nChange-Id: I78915c332169b26ffb2b97310efedec65bc25e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/12f06f7858b44bc79b1d023c1d98cdfa193d5bcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/12f06f7858b44bc79b1d023c1d98cdfa193d5bcf"}]},"branch":"refs/heads/master"},"b0d4d34e7f272bac0bcc2f74c73ae8c18688d75f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-04 16:24:58.000000000","uploader":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"ref":"refs/changes/11/840511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/840511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/840511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/840511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/840511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/840511/2"}}},"commit":{"parents":[{"commit":"a5405554577f16c7d498ef55844d00b1fce6aa9d","subject":"Merge \"[SVf]:Fix multiple lsvdisk calls for GMCV create volume operation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a5405554577f16c7d498ef55844d00b1fce6aa9d"}]}],"author":{"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","date":"2022-05-04 15:29:51.000000000","tz":0},"committer":{"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","date":"2022-05-04 16:24:58.000000000","tz":0},"subject":"Volume Deletion rejection message  does not include \"awaiting-transfer\" in invalid volume state","message":"Volume Deletion rejection message  does not include \"awaiting-transfer\"\nin invalid volume state\n\nWhen user tries to delete a volume in \"awaiting-transfer\" state,\nthe error message received does not include the \"awaiting-transfer\"\nstate in invalid states of Volume for deletion.\nThis leads to the user believing the volume-delete request is valid\nand is not able to debug the reason for failure of the volume deletion\nfor volume present in \"awaiting-transfer\" state.\n\nCloses bug: Bug #1971603\n\nChange-Id: I78915c332169b26ffb2b97310efedec65bc25e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b0d4d34e7f272bac0bcc2f74c73ae8c18688d75f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b0d4d34e7f272bac0bcc2f74c73ae8c18688d75f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6aad0962839c10a2158c360abbabe299fb6020a7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-05-04 16:32:03.000000000","uploader":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"ref":"refs/changes/11/840511/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/840511/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/840511/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/840511/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/840511/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/840511/3"}}},"commit":{"parents":[{"commit":"7e04b5b0d4c91df7601e8e7f36d49fb6d651ea2b","subject":"Handle the case when tempest fails","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e04b5b0d4c91df7601e8e7f36d49fb6d651ea2b"}]}],"author":{"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","date":"2022-05-04 15:29:51.000000000","tz":0},"committer":{"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","date":"2022-05-04 16:32:03.000000000","tz":0},"subject":"Volume Deletion rejection message  does not include \"awaiting-transfer\" in invalid volume state","message":"Volume Deletion rejection message  does not include \"awaiting-transfer\"\nin invalid volume state\n\nWhen user tries to delete a volume in \"awaiting-transfer\" state,\nthe error message received does not include the \"awaiting-transfer\"\nstate in invalid states of Volume for deletion.\nThis leads to the user believing the volume-delete request is valid\nand is not able to debug the reason for failure of the volume deletion\nfor volume present in \"awaiting-transfer\" state.\n\nCloses bug: Bug #1971603\n\nChange-Id: I78915c332169b26ffb2b97310efedec65bc25e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6aad0962839c10a2158c360abbabe299fb6020a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6aad0962839c10a2158c360abbabe299fb6020a7"}]},"branch":"refs/heads/master","description":"Rebase"},"89f5311e50a7a8540a9de70365a04db7ac28962b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-05-04 17:09:11.000000000","uploader":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"ref":"refs/changes/11/840511/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/840511/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/840511/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/840511/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/840511/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/840511/4"}}},"commit":{"parents":[{"commit":"a5405554577f16c7d498ef55844d00b1fce6aa9d","subject":"Merge \"[SVf]:Fix multiple lsvdisk calls for GMCV create volume operation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a5405554577f16c7d498ef55844d00b1fce6aa9d"}]}],"author":{"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","date":"2022-05-04 15:29:51.000000000","tz":0},"committer":{"name":"b-prajakta","email":"prajaktabelapurkar11@gmail.com","date":"2022-05-04 17:07:08.000000000","tz":0},"subject":"Volume Deletion rejection message  does not include \"awaiting-transfer\" in invalid volume state","message":"Volume Deletion rejection message  does not include \"awaiting-transfer\"\nin invalid volume state\n\nWhen user tries to delete a volume in \"awaiting-transfer\" state,\nthe error message received does not include the \"awaiting-transfer\"\nstate in invalid states of Volume for deletion.\nThis leads to the user believing the volume-delete request is valid\nand is not able to debug the reason for failure of the volume deletion\nfor volume present in \"awaiting-transfer\" state.\n\nCloses bug: Bug #1971603\n\nChange-Id: I78915c332169b26ffb2b97310efedec65bc25e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/89f5311e50a7a8540a9de70365a04db7ac28962b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/89f5311e50a7a8540a9de70365a04db7ac28962b"}]},"branch":"refs/heads/master"},"2b75fa3021e1e6b4d635ab36a9084719c0e3d969":{"kind":"REWORK","_number":5,"created":"2022-05-05 17:17:28.000000000","uploader":{"_account_id":34807,"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","username":"b-prajakta"},"ref":"refs/changes/11/840511/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/11/840511/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/840511/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/840511/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/11/840511/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/11/840511/5"}}},"commit":{"parents":[{"commit":"a5405554577f16c7d498ef55844d00b1fce6aa9d","subject":"Merge \"[SVf]:Fix multiple lsvdisk calls for GMCV create volume operation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a5405554577f16c7d498ef55844d00b1fce6aa9d"}]}],"author":{"name":"Prajakta Swapnil Belapurkar","email":"prajaktabelapurkar11@gmail.com","date":"2022-05-04 15:29:51.000000000","tz":0},"committer":{"name":"b-prajakta","email":"prajaktabelapurkar11@gmail.com","date":"2022-05-05 16:57:15.000000000","tz":0},"subject":"Update Volume Delete rejection message","message":"Update Volume Delete rejection message\n\nWhen a user tries to delete a volume in \"awaiting-transfer\" state ,\nthe error message received does not include the \"awaiting-transfer\"\nstate in invalid states of Volume for deletion.\nThis leads to the user believing the volume-delete request is valid\nand is not able to debug the reason for failure of the volume deletion\nfor volume present in \"awaiting-transfer\" state.\n\nCloses-Bug: #1971603\n\nChange-Id: I78915c332169b26ffb2b97310efedec65bc25e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b75fa3021e1e6b4d635ab36a9084719c0e3d969"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b75fa3021e1e6b4d635ab36a9084719c0e3d969"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
