)]}'
{"id":"openstack%2Fcinder~841315","triplet_id":"openstack%2Fcinder~master~Ide947c8e3d3f050665c3acb804b6617d2b9c045c","project":"openstack/cinder","branch":"master","topic":"mypy","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-01-02 18:48:24.000000000","reason":"Change was submitted"},"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2026-01-02 18:48:24.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-01-02 18:48:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ide947c8e3d3f050665c3acb804b6617d2b9c045c","subject":"mypy: NFS driver","status":"MERGED","created":"2022-05-10 18:55:59.000000000","updated":"2026-01-03 01:55:56.000000000","submitted":"2026-01-02 18:48:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841315-mypy","meta_rev_id":"28c523bf9aaf4bd402b5780ee468644b9b344340","_number":841315,"virtual_id_number":841315,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-02 18:48:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-01-02 16:37:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2025-12-19 16:58:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2026-01-02 16:37:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-10 18:56:09.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-05-10 18:56:14.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-10 19:21:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-10 20:10:25.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-05-10 20:36:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-10 20:38:23.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-10 21:19:36.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-05-10 21:26:21.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-10 22:21:24.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-05-10 22:49:29.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-11 00:12:53.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-05-11 07:36:09.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-11 14:23:16.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-05-12 10:38:52.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-27 14:36:28.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-07-27 16:11:23.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-27 17:19:09.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-07-27 18:32:21.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-27 19:43:21.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-27 19:58:51.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-07-28 06:13:23.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-07-28 11:51:54.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2025-03-21 13:37:32.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-03-21 13:38:21.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-03-21 20:49:12.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-03-22 09:12:01.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-27 21:08:43.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-04-03 10:49:14.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2025-04-03 16:21:42.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2025-04-03 20:47:44.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2025-04-04 01:12:31.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2025-10-09 19:55:19.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2025-10-10 11:54:08.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"CC"},{"updated":"2025-11-11 00:37:21.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2025-11-25 14:44:00.000000000","updated_by":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"reviewer":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"state":"CC"},{"updated":"2025-12-19 16:58:39.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-12-19 20:55:52.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-01-02 16:37:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-01-02 19:35:59.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-01-03 00:43:57.000000000","updated_by":{"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},"reviewer":{"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},"state":"CC"},{"updated":"2026-01-03 01:55:56.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"}],"messages":[{"id":"89f48a31a7c1872113bf3b09aa1551c6ab120b69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-10 18:55:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"349f7802c8c907f8ae745e3a3092e374baef65e5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-10 18:56:09.000000000","message":"Patch Set 1:\n\nBuild Started","accounts_in_message":[],"_revision_number":1},{"id":"ca60550520872c2109a14648aca23547694941f2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-10 18:56:14.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15976/","accounts_in_message":[],"_revision_number":1},{"id":"36bf25ef4c4e610e95aa058b2c5e42a7afa30f04","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-10 18:57:55.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/15976/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"90dd6a495b749ce11c229f5968eb337140e9fc99","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-10 19:16:27.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1652208969086 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"1e9f13bc55919d52b99598d6fe7008669eb6cca2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 19:21:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b9a46781c9b42178df9e5e6a4634006 : SUCCESS in 21m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9905fc9eef3c463ca886e02dd55e3aee : SUCCESS in 21m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b7909ec1aff728561f29ed1e87f6b9b04ab751a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-10 20:10:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/841315/1/iscsi/ : SUCCESS in 1h 13m 31s\n- cinder-smb http://cloudbase-ci.com/cinder/841315/1/smb/ : SUCCESS in 1h 08m 08s","accounts_in_message":[],"_revision_number":1},{"id":"6ea9bdce079d36a6cfbae08c84a60aa0cdb22b02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 20:36:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/664a5be320024bfd818c9257898c87ab : FAILURE in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58f3c9745af143e5b2b84b0873a9f682 : FAILURE in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a84b83be709849ababc11a24fcadaad6 : SUCCESS in 11m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa3096579bb4466da29f1cff5dc2f4bf : SUCCESS in 11m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d073e527c0a948af8ec86bd46b92ea6e : FAILURE in 5m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/65c45235968d434aa09a6b26950cef03 : FAILURE in 19m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6f97047371c34ae19bd6313b9aa53b2f : SUCCESS in 1h 16m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16672532e0b74b2bb8ced7730f1bff24 : POST_FAILURE in 22m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4dfae40592614d1f80af80e9bb980604 : SUCCESS in 15m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1677a5cc871f42c38ea1703ba62b4052 : FAILURE in 3m 47s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0e5fc2c6b85547b99ea072a2f03064b8 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/df5a6c29af5043829fd6df8663456ae6 : FAILURE in 3m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/131027efb0564f66971dc6cedf8bccaf : SUCCESS in 6m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/62e42e37db8c4888a6c32ee8bf8f797a : FAILURE in 19m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a6ab320cc0b04348aae5de1cd96801a3 : FAILURE in 7m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f63a65f8c5b04156ad59e26dad3855c0 : POST_FAILURE in 20m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/21d9c1671b204b8abc44f484264a89a3 : FAILURE in 1h 32m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b0c9413e3b2d4475bc9e81ed8bf7ee07 : POST_FAILURE in 18m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/680910a89cf94d37861eb8bf0996293d : FAILURE in 26m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3a72515002054e42a1a6a5d35c3e9e61 : SUCCESS in 1h 27m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/629bd69507394a169c94b32205ab1db3 : SUCCESS in 50m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6b1a92a7cd40444792ecd626aca6393f : SUCCESS in 1h 26m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2cbdabfcf7c74957b2daf1109b3e2132 : FAILURE in 22m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da6b3aec16704798b7a4f45c3f615aa2 : SUCCESS in 57m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ecfb4cb7f67204503fdab741891674d91445d7b0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-10 20:38:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/841315 : FAILURE in 1h 02m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c3b00f93e35889f6c5d2d7d9c1707d54c5b6d009","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-10 21:19:36.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"d381207699987a54a7fce3c50c4d68787d211a97","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-10 21:26:21.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/841315/1/check/tempest-dsvm-full-focal-py3/f2eabc1/ : SUCCESS in 1h 49m 12s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/15/841315/1/check/devstack-plugin-ceph-dsvm-focal-py3/a69d14d/ : SUCCESS in 2h 20m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e5653c1f6a2da06c26789d605be33d98cd5a8795","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-10 22:21:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/15/841315/1/check/iscsi-driver-nimble/5e015e7/5e015e7.html : FAILURE in 1h 00m 31s","accounts_in_message":[],"_revision_number":1},{"id":"90f3d6167e6cc2d89382f683b3fbccd1491c21af","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-10 22:49:29.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-15-841315-1/ns5_nfs : SUCCESS in 3h 39m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-15-841315-1/ns5_iscsi : SUCCESS in 3h 24m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"56d6fcc636e2fa41d3b3872bbdf25c0d432cdaeb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-11 00:12:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/15/841315/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/ef70862/ : TIMED_OUT in 3h 02m 46s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/15/841315/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/5c72bca/ : SUCCESS in 1h 27m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1a9a8f84e7cd5468ebf04360ba65082ad1c6349c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-11 07:36:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b9c908724c1f4214aac5cb066d91ac7d : SUCCESS in 1h 34m 16s","accounts_in_message":[],"_revision_number":1},{"id":"eb80120cf4a667257b5904fd974d6bf2eaf4178d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-11 14:23:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/15/841315/1/20220511205608/NEC-V-FC : SUCCESS in 1h 49m 44s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/15/841315/1/20220511205608/NEC-V-ISCSI : SUCCESS in 2h 27m 05s","accounts_in_message":[],"_revision_number":1},{"id":"43b55c28c8d583a8485fdcc7c638840b2db22646","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-05-12 10:38:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/15/841315/1/20220512161807/NEC-ISCSI : SUCCESS in 3h 19m 51s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/15/841315/1/20220512161807/NEC-FC : SUCCESS in 3h 20m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e06b6fc933c2bae1b4023781ab0bc404b9e0aeb8","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-05-16 18:53:20.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"0b487d9944d4a44d27813c582e3ad8c59107b215","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-07-27 14:36:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c4acee6c943cdffe7a2237e84676bfd831ab968f","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-07-27 14:36:27.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"7c9ed8722c6de85aa908ce5303191f54b81a3338","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-27 14:36:28.000000000","message":"Patch Set 2:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"f81185d36a6f0787cdc62fc77aa98fc71f6f4c5a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-27 14:39:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"c73953c8eba03ed6cfeeff9f297ce062213a1560","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 15:04:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93ae39ef1a274fadab151f2fdff1da41\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a004e201f2f9485381885a8aff6145b1 : SUCCESS in 16m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be7bc51ddf20425f888c5341b896b621 : SUCCESS in 25m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"60bdbba7c9e160cefba359d412070469bb944224","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-27 15:11:52.000000000","message":"Patch Set 2:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/289 \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"2a53ab08ac5ebd99ff91ecd9c281c1d16e2fec70","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-27 15:19:24.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1658932587918 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"0bfad6861d608e875f1956a71bb90c8fc8e18eb9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-27 15:41:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/15/841315/2/check/iscsi-driver-alletra-6k/ecf8a28/ecf8a28.html : FAILURE in 1h 03m 53s","accounts_in_message":[],"_revision_number":2},{"id":"34c2de0b40cc3303bd7e0ec083fc57e4f2e954ae","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-27 15:45:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/841315/2/iscsi/ : SUCCESS in 1h 05m 29s\n- cinder-smb http://cloudbase-ci.com/cinder/841315/2/smb/ : SUCCESS in 1h 08m 08s","accounts_in_message":[],"_revision_number":2},{"id":"c33f66216aebad9a3b9986e133b9d4d1e572670f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-27 16:11:23.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/15/841315/2/check/fujitsu-eternus-dx-iscsi/13ffeb9 : SUCCESS in 1h 34m 20s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/15/841315/2/check/fujitsu-eternus-dx-fc/54da570 : FAILURE in 1h 34m 00s","accounts_in_message":[],"_revision_number":2},{"id":"40fdae951c55c354053bbb6804c81d6ebb3a97c2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-27 16:20:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/15/841315/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2372a47/ : SUCCESS in 1h 43m 01s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/15/841315/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ff0c71a/ : SUCCESS in 1h 25m 28s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/15/841315/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/ebf949b/ : FAILURE in 33m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e5b228b4d0899b65cdd18c2c6deafd820efdce4a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-27 17:02:09.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/1c5b633df2b04a67ab5edfdd1d38d7c6 : POST_FAILURE in 1h 43m 08s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/ecfae00cd25042a6822787bf9efb47c8 : POST_FAILURE in 1h 59m 46s","accounts_in_message":[],"_revision_number":2},{"id":"a57b8d5b5b9e44486ea501103f7823af11a0bffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 17:04:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15cd1e49a2324fba93aec3c378891cfa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c240b4ad24cb4337987bf613ae21781c : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/416f059bffbe42c6beba6a5ba8ab5bb9 : SUCCESS in 14m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09b7464d81cd45d0990a999031144b43 : SUCCESS in 14m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a640e19f97d74820877a2e79cfeb9406 : SUCCESS in 12m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/887899e442f140ee99ef4800988bfe85 : SUCCESS in 12m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/8a7a44c14c48474d9fd00a49f30fd3f5 : SUCCESS in 55m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e7cd9c6a38d14445a23eb442abba8b3f : SUCCESS in 1h 07m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c44d902626884ff2a2ad47820b348f2a : SUCCESS in 49m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3d6f7571ecf1489bb99d500b16618a39 : SUCCESS in 15m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f00fb317e75f43798c2efe59ef4b9bad : SUCCESS in 4m 40s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8859de511ec542b4b21d9a45da0349fb : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fffc37e7c88045049f0a4e34072d1cde : SUCCESS in 7m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8e8327396de1482287da67ef17f6a407 : SUCCESS in 8m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9945bd0fa6e74537b59cedb89461016b : SUCCESS in 37m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7ef4385ee8494c01a7150c5ed1e45ec8 : FAILURE in 7m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/61e901e52b594cc3b782b0f90f73db4c : SUCCESS in 1h 25m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/da048d0f42cb49f5bfdc82fda75ff02a : FAILURE in 1h 22m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6e85968a34834c26b093210091bbc644 : SUCCESS in 1h 31m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e741851fc2fb41458a1927cebbb6648f : SUCCESS in 1h 25m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f07589b936534e82b98657864339da90 : SUCCESS in 1h 32m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/373647069d76457cbfc23a81e694b3ae : SUCCESS in 42m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/84a872f0a7f64d10ade90750377f312a : SUCCESS in 1h 33m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9004cb8efeee41818b9739b79b5e5d42 : SUCCESS in 2h 23m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0346a3006118455d8ebfe7ec51aa61af : SUCCESS in 50m 35s","accounts_in_message":[],"_revision_number":2},{"id":"4ba5c8cf44aa2eed720570a5a239df616e931131","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-07-27 17:19:09.000000000","message":"Patch Set 2:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/27_07_2022_17_11_05\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"f1890e25ca5f65b7d3ae88ff00e521dffa4529d5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-27 17:33:51.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/841315 : FAILURE in 2h 55m 27s","accounts_in_message":[],"_revision_number":2},{"id":"f29fce4fa25ca70f39cdd632dc66c4f611036efc","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-27 18:05:44.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-15-841315-2/ns5_nfs : SUCCESS in 3h 25m 43s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-15-841315-2/ns5_iscsi : SUCCESS in 3h 28m 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"e75149790b8b73035bd9b56dc88083d6e1b160c8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-27 18:32:21.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/15/841315/2/check/DellEMC_PowerFlex/b3578a4/DellEMC_PowerFlex/1087 : SUCCESS in 1h 27m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"966bcc3da3f07e120565f869a7de1f365dbf5b8d","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-07-27 19:43:21.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/15/841315/2/check/DellEMC_PowerMAX_FC/456e425/DellEMC_PowerMAX_FC/760 : FAILURE in 42m 32s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/15/841315/2/check/DellEMC_PowerMAX_iSCSI/2ea9997/DellEMC_PowerMAX_iSCSI/743 : FAILURE in 48m 41s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"8858bd39ad040bce6c449b2f15de2123b82af37a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-27 19:58:51.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/15/841315/2/check/DellEMC_PowerStore_NVMeTCP/c5de58d/DellEMC_PowerStore_NVMeTCP/42 : FAILURE in 2h 53m 36s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"bcfea9a5192c44c2aa918614215540d548706a98","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-28 01:25:11.000000000","message":"Patch Set 2:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":2},{"id":"d5aec22217cfa6e9e67d39066fffee1d331f493e","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-07-28 06:13:23.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/15/841315/2/check/DellEMC_SC_FC/2a465d6/DellEMC_SC_FC/819 : SUCCESS in 55m 22s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/15/841315/2/check/DellEMC_SC_iSCSI/53753d5/DellEMC_SC_iSCSI/839 : SUCCESS in 1h 51m 26s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"e702810568a9da9653b8470cdb6109be2362d8f4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-28 11:51:54.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/15/841315/2/check/DellEMC_Unity_FC/360d097/DellEMC_Unity_FC/1280 : FAILURE in 19m 16s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/15/841315/2/check/DellEMC_Unity_iSCSI/02e51a0/DellEMC_Unity_iSCSI/1173 : SUCCESS in 2h 03m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"db1a3efa26c596fe1f7ea633ba79cba92eae1132","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-30 01:09:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/15/841315/2/20220730075358/NEC-V-ISCSI : SUCCESS in 2h 06m 43s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/15/841315/2/20220730075358/NEC-V-FC : SUCCESS in 2h 15m 21s","accounts_in_message":[],"_revision_number":2},{"id":"26dcee81037cf8bde37f26655f701a97af12db61","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-06 00:17:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/15/841315/2/20220806054909/NEC-ISCSI : SUCCESS in 3h 28m 24s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/15/841315/2/20220806054909/NEC-FC : SUCCESS in 3h 28m 34s","accounts_in_message":[],"_revision_number":2},{"id":"e4654ea700d73e794b6c3f3e4c82c0f1bad2c909","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-03-21 13:37:24.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e1f22ce5fb1541278ac289c952785976c41d993a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-21 13:37:32.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1552/","accounts_in_message":[],"_revision_number":3},{"id":"1eb315f7362617eb737547d402994c77197b0c57","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-21 13:37:34.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/841315/3/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/841315/3/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"ad298093c907abbc2a2ad91e85d3195acf0e7bd8","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-21 13:38:21.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"83a660d35ef2d08fe652e7a5f292944dde1bc870","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 13:53:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5870e596de44eb784b75e16aa945002\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6c3acb7d7a9a4e1bbaacc98899bc27d7 : SUCCESS in 12m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/51c22becd1264358b4a6315a337fd186 : SUCCESS in 14m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2a74e9ec4e5098beef9413902a2b41dbbd596373","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-03-21 14:13:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7abc97fe7e80ff4fa8587ed1656595cc48a797c4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-21 14:13:22.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1553/","accounts_in_message":[],"_revision_number":4},{"id":"1e47bfcca75d6d8769fa03216c950530b19419a4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-21 14:13:24.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/841315/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/841315/4/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"637f03aae4ad3fe33beccd62fafb7b55c44b03ae","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-21 14:14:06.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"543e735b4a10e987b402a6f75017105e864fb271","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 14:28:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32a583540d2a4ff7a6b2e59b41d9f157\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/95001c203efd471ba8808f1b084ec53a : FAILURE in 8m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c8a2e8d3a96041d780405011cc71a88c : SUCCESS in 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7156f19be192f81c6649b047f5bbfb2ded64cd13","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-21 15:05:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/492/.","accounts_in_message":[],"_revision_number":4},{"id":"e02b0ce87f4cc0eadbcc9170cb5d766225c91f23","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-21 16:13:20.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/558/.","accounts_in_message":[],"_revision_number":4},{"id":"145432de2ef5f7631d0ad098250c10068b120800","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-21 16:25:51.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9b5bd0eb33a24ce4bed5cb0ea3511e5f\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d599a5b2281b40feb8c1f94bae1e7888 : SUCCESS in 45m 37s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f08bbd1e824b452cb49a0e5224f06554 : FAILURE in 1h 58m 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/298fa35fdc0047d082ace3287ffcac35 : FAILURE in 2h 12m 08s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fd683267b0e143439738146516d9cfb7 : SUCCESS in 42m 44s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/609366f5cfd842959ea4e7bca56a5708 : RETRY_LIMIT in 27s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bab4bc5522384f8ebbff77b8643ce1fd : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/16967d7ec8c84fb3b0a2e0d1b66d2327 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/400f2242d878485aba15ed05291f5cda : RETRY_LIMIT in 26s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob ... (truncated)","accounts_in_message":[],"_revision_number":4},{"id":"1f3f04072f100d99c542a34a74ad9768a67c7166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 17:22:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f93fa309c4be4c79b555e11fb636a00f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c5f9b9f9295473ab8b1a6fb1df556f3 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/413903cd7bd44072bb46edc2bb2fe045 : FAILURE in 6m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b96ab2f36a44ac7b39014c8bbe581ad : SUCCESS in 12m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/016b91cfc93243b5b678ed1d5985aa75 : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/5c04edd79dda4c4099f6f1c979658d8f : SUCCESS in 52m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0f4609a6c6724040922952c9437b525e : SUCCESS in 30m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7b89acf803304274a42de1d631b8cf58 : SUCCESS in 1h 17m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b37c0be79cdc476589d8c06ebc397620 : SUCCESS in 54m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a89d2395602644c8a772f82ba2e3de0b : SUCCESS in 23m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0f4d026bc229466db32205b10d3f3175 : SUCCESS in 6m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ba6f117a421d481484e70cef52ea9ac6 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/abe3f705e56341d59d12d74e08314a90 : SUCCESS in 6m 09s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3afb2197466d46a1a243b3af022929d2 : SUCCESS in 5m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c7e85d1bfb8742b28d129170cefe2482 : FAILURE in 21m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/672feac561064399aeb8050c7cd8ad3a : SUCCESS in 4m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/70e97836057a4c9d8e30e0a70bafdf3f : SUCCESS in 1h 42m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b568e9c13028400196b0372ac16401e1 : TIMED_OUT in 3h 07m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/57f194d5f5794f7eb6712cb72152725b : SUCCESS in 1h 49m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/135b13eada3f436e9ef5c2731526ff55 : SUCCESS in 2h 02m 43s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/74698814e6b84841a767c316c1ed1c6c : SUCCESS in 36m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4710f2d4dfb740dfa226e52bd0b855a4 : SUCCESS in 56m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8bf2aea08ec04d3686ddc1a1a2d0460f : SUCCESS in 1h 25m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b27ade01985c4359a68b03b04ab35099 : SUCCESS in 23m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6c5ae9e9950f42d9a81ced8010436f0c : SUCCESS in 1h 28m 39s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8fbebb29928a4acfbc229dda2845eaf0 : FAILURE in 22m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/93f3178b3ac646658d441f969ab576e7 : SUCCESS in 2h 00m 33s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae91630b52854ba489bf2af13b508779 : SUCCESS in 1h 32m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aead1ecb24f24c20b8b730b9d91492ea : SUCCESS in 27m 44s","accounts_in_message":[],"_revision_number":4},{"id":"ef9b2f067de8ed50472df90946b8aa7a49dd8d02","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-21 17:29:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/841315 : SUCCESS in 42m 41s","accounts_in_message":[],"_revision_number":4},{"id":"cbb1d848acfd922a4b70d90203c67966de5afae1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-21 17:43:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/841315 : SUCCESS in 59m 14s","accounts_in_message":[],"_revision_number":4},{"id":"3624bed6c107a4bad651727c8516d90eeaea66fb","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-21 20:49:12.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-841315-3/FC/ : SUCCESS in 2h 15m 57s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-841315-3/iSCSI/ : FAILURE in 7h 11m 20s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"c982a2056f4612382580bf86603dce9ce2022b20","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-21 23:27:48.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-841315-4/FC/ : FAILURE in 9h 14m 8s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-841315-4/iSCSI/ : UNSTABLE in 34s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"41a4f3c4ad68bd3d0f08f4628f30252121263a3c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-22 05:36:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4a7723ea24fe4482957f54f52a11bd01 : SUCCESS in 1h 16m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/38a8f5d9d4a449dd85b82f81d033b0d3 : SUCCESS in 1h 15m 35s","accounts_in_message":[],"_revision_number":4},{"id":"30096cdee930570793d6dfbb1d64acfca44832e7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-22 09:12:01.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"1d2516a6cd80354c1249852e89e91b52b9159a2a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-03-25 13:50:42.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"bf8d22edd10c260a627b87027c54d94874354220","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-25 13:50:53.000000000","message":"Patch Set 5:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1560/","accounts_in_message":[],"_revision_number":5},{"id":"d5ded97a06fbe230225b93a6993b6601cfac7547","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-25 13:50:55.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/841315/5/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/841315/5/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":5},{"id":"5718bddc7b9713835ddb271488c1e9b51b9b85a6","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-25 13:51:48.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"dcb8a398b232b6d10d8b30abef8916fe0016e5fd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-25 13:51:54.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"c08673f430f616df35cbe7e45ce7f059a0ecad60","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-25 14:26:57.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/841315 : FAILURE in 34m 33s","accounts_in_message":[],"_revision_number":5},{"id":"792cb774f83f5621985ca9131a8b79233b2f8aba","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-25 14:47:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/497/.","accounts_in_message":[],"_revision_number":5},{"id":"79bcfcba1019c1ea6733a6badac264b5253cbd6d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-25 15:08:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/563/.","accounts_in_message":[],"_revision_number":5},{"id":"1a098a07ccee16736b67fc89281cc761c5fb315e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-25 15:12:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d0f033cce1124826b4efd241f4c50ce0 : SUCCESS in 1h 18m 08s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/67891c26111947bf9344511962fb20ef : SUCCESS in 1h 20m 02s","accounts_in_message":[],"_revision_number":5},{"id":"e4ea371512de2e93971d4df4d6810556881c0004","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-25 15:24:56.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-841315-5/FC/ : FAILURE in 1h 33m 52s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-841315-5/iSCSI/ : UNSTABLE in 30s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"44d852146d0a7ef5c41240832b330ba0f33f6704","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 15:24:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e67614dae9a496094206751fea6a5a0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0634d9f6e25846e1b7516e6b89407eeb : FAILURE in 9m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eade536f1cde4457b60561cb1d3f924d : SUCCESS in 16m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2c0e663f8e58e99f874b194c0d20a6425150a009","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-25 15:28:07.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"93fc6c856a1cf43d9eb9a393b2ec8b140b0e1f92","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-25 15:32:21.000000000","message":"Patch Set 5:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a9ace27c7d4e4b899ac59a37c5f07701\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/04963519977a4cf4851dfbef0f637823 : SUCCESS in 45m 13s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/34627de00c654c92bcfbbf87ab930ab3 : SUCCESS in 57m 12s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1b13290b1ed54bafa6fc0e5edbb7c135 : FAILURE in 1h 41m 01s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/85ef6598850c4782b29b9813bb16f6d9 : SUCCESS in 38m 15s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cd1f8c825a804fbf86dccc22e28cbd5c : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e9681e12afb6493488530da7fbcbf0b1 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f1caef975bf74c73a3334a7b912ac5c2 : RETRY_LIMIT in 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9db60c7e33c243ec845d1b6785acfa5a : RETRY_LIMIT in 27s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":5},{"id":"65cfff6428fd4318a77ebf2a4e3d8bf7cb1839fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 15:56:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/436735748dfe4249b1e64795e25efa26\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d793cbd4eed54f96a1286646c64dcda5 : FAILURE in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a9305d4244246b8b4f5ed2684ba85ba : FAILURE in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b05edebace18403498bbbd04dabf37ed : SUCCESS in 12m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49782427242746b88dcaadf58500a076 : SUCCESS in 16m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/5cb13456742642e08a46d3667bd8e5c0 : SUCCESS in 49m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cd4ea4e0101246a9bf2c9e40509d8f85 : SUCCESS in 52m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8874a510f48f419b86a94dfe6c04719f : SUCCESS in 1h 22m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7d10c4b4e294015b53e636f652b9396 : SUCCESS in 1h 01m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bc3b067146a049f7966a5e4cd7eefa65 : SUCCESS in 23m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7f8d5edcc56446cd880313e2ed1569b0 : SUCCESS in 6m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5a100747f8bb4c59b24e8ea41111cf7c : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2b7b728d14e7420ab993220ba0e87480 : SUCCESS in 6m 26s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/59627eb81b1347b8b7f28a7e3064f244 : SUCCESS in 5m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8462679588704b33bd5e6853e1babb0a : FAILURE in 29m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/efa79fbd89fa4841b49447b35e38703a : SUCCESS in 26m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/57d78c57d0a24441be0d50a96eac30b5 : SUCCESS in 1h 46m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d2c0b46c59e045dab28da46953d37ecb : FAILURE in 1h 41m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/25fae3f0555d4b0c85580a74a42e2a01 : SUCCESS in 1h 49m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/dccce4e9cd274d7ebefdc5cb839bf8e3 : SUCCESS in 1h 48m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/52b04a999436401cb1294752e7bff57d : SUCCESS in 33m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/515985d17c7146808b96a9b67e83cc6b : SUCCESS in 1h 02m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a2e3bb7a6ed84c3d927a4b6c1a13764b : SUCCESS in 1h 43m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c70985d5fd994e058932125307ac5ab6 : SUCCESS in 27m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1be43622b1e746e682f126688d1ea45a : FAILURE in 1h 10m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1e6ac112720d4eb6a0a07c8765417db8 : FAILURE in 32m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/369f289a6fe64dd1ac844fa0ed3ad50c : SUCCESS in 49m 57s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/942200d86a3c4133b3aecbec0e362037 : SUCCESS in 1h 33m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2763dcc1c6ad48e2baa681cb483f10f7 : SUCCESS in 49m 22s\n\nWarning:\n  Comments left for invalid file cinder/tests/unit/volume/drivers/test_nfs.py","accounts_in_message":[],"_revision_number":5},{"id":"54101ae45409a7c8160dfcc6459ae6583503e340","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-03-27 14:54:49.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"5ee204007e2e5bf4720948b945169429df3b3925","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-27 14:54:58.000000000","message":"Patch Set 6:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1606/","accounts_in_message":[],"_revision_number":6},{"id":"dcfe85508fd6c5e23d7554aba23dfbac019fafba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-27 14:55:01.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/841315/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/841315/6/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"5445df340d25850f3c07aa8630631ec4fb76cad6","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-27 14:56:06.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"5328a83f481455b282106fb5e58adf14439d7fc2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-03-27 15:09:58.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"83600f6728b3b6831614b2c77189fa22b4beb520","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-27 15:10:08.000000000","message":"Patch Set 7:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1607/","accounts_in_message":[],"_revision_number":7},{"id":"6c5724ea4cc9077d029d85f87269d3ed14122719","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-27 15:10:11.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/841315/7/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/841315/7/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"2e14503910b7850de0622cebeb2aba6f03512af7","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-27 15:11:11.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"c4fb8f1eaf5fb03f3c3cd9238dfd8172c22f8e51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 16:03:11.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81569c24b7a241ff853c2927ae7bebd1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b9c0b786ac13437587e4e381fd48d005 : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/33466cab30f34040ad1d4b9e25c4044c : SUCCESS in 12m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9bc737422d741dc70dd2a5fd1ea251e41eac7064","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-27 16:10:42.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/841315 : FAILURE in 33m 22s","accounts_in_message":[],"_revision_number":7},{"id":"0b933df99f0b25c9d7357bd7749d6cf275510a50","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-27 16:18:41.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/841315 : FAILURE in 35m 40s","accounts_in_message":[],"_revision_number":7},{"id":"59e043ef1e9b359b4160bffac8aa3360660314c9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-27 16:39:24.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/510/.","accounts_in_message":[],"_revision_number":7},{"id":"eb505b569f38e412bf17841f4c642aa48505e87c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-27 16:39:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/595/.","accounts_in_message":[],"_revision_number":7},{"id":"f88d458f8637358578c4c06d99aef9d5818fa736","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-27 16:53:23.000000000","message":"Patch Set 7:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ea86e66214004c3c8c49afff65c5fdde\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/979f9d3e834a4b64adcdf14ff2d7390e : SUCCESS in 44m 03s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d4b8fd4647b74a23a121777ecdfa9404 : SUCCESS in 49m 52s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0cfdcaad0d75400a807af2115ff0345e : FAILURE in 1h 42m 36s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/494873c93e2843c582ece200f7898122 : SUCCESS in 40m 54s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b60177bc99be4a2586925bd7eedacdf9 : RETRY_LIMIT in 51s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8cc723e7b54f4bb0b298aaa181670116 : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/77ffaecf864143ddb096abe309fd1d66 : RETRY_LIMIT in 29s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5816f9cbf2fd44fc9732f1e7a0cb8577 : RETRY_LIMIT in 29s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#512\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#469\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":7},{"id":"c84ffc86fbed63b364e15dee6e6015e263f458a6","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-27 17:26:47.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-841315-6/FC/ : FAILURE in 2h 31m 31s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-841315-6/iSCSI/ : UNSTABLE in 36s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"efcd595896978374ab49d718837f588618d90b72","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-27 17:28:36.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-841315-7/FC/ : UNSTABLE in 34s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-841315-7/iSCSI/ : FAILURE in 2h 18m 15s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"575bce6c38273ac965eee70d099be42278729d80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 17:46:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3b767fcbe454f86b4986fb32a5a8d86\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f57fe04b1c54b36afb54badf17f25be : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c3d66a1de0e4c21ad4f06295b8fa2a9 : SUCCESS in 12m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1c32a8d0d74a412cb87e132314a2819f : SUCCESS in 12m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/498e24ce52f74ba7b0076a1fe12f9322 : SUCCESS in 13m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/8f43196066dc4192aeac4aa8415fd626 : SUCCESS in 50m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f1aa235751e544568326cf846db53155 : SUCCESS in 52m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/be2cfe0c337f4ef79ed8e5c962c2f670 : SUCCESS in 1h 26m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97a3b30f55774d1a8b3c514c97ceb30b : SUCCESS in 53m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4a77107239454f22948ffa812a292c6a : SUCCESS in 21m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f42ddfd672ff4fa1903d4a4fc38f7f4c : SUCCESS in 6m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d360144636204f4ab9dc06429cb226ee : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/10d5e9fc3bde4a8e9941d2e8ef245494 : SUCCESS in 7m 10s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a88de2de3310470c8f41d4987f0cab21 : SUCCESS in 6m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1c95ae3e6e83471f9353757fc7e93f9f : FAILURE in 23m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/25630281ed614ff899f9a4ee664b3d21 : SUCCESS in 7m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/28383e35698b4f63b59f4a2984981fa2 : SUCCESS in 58m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5eb7143029f042d2974f8341477cf327 : SUCCESS in 2h 27m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a977d1b86129450d8715593ac6213e0f : SUCCESS in 1h 55m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/23264fadd568405085b98da2aa5d1695 : SUCCESS in 1h 46m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ef833aa4f31c4b2ab97cad1720455eb1 : SUCCESS in 33m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/84c0486eb0c144e7acc69f1ad4b0295e : SUCCESS in 30m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6823f2ccd8f44162b49e2cc680c7be93 : SUCCESS in 1h 37m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f46b833946c54919a2e8227a5fd938bf : SUCCESS in 23m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0477b71840f84449923fd5ddddc60649 : SUCCESS in 1h 36m 06s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/843dcd68c73f4872965f9dcf487fde0b : FAILURE in 1h 04m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/27b4b0c162ee4bd4af61215da37b7f88 : SUCCESS in 1h 48m 26s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8ff92d1a7f346b0a3fdc8010e417a06 : SUCCESS in 1h 48m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a645f93370b245198acf5e37ecf55e8c : SUCCESS in 56m 32s","accounts_in_message":[],"_revision_number":7},{"id":"25eaec91be675509c53042b5f50665b02d6a859d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-03-27 18:17:09.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/5/20250327192000/NEC-V-FC : SUCCESS in 4h 17m 32s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/5/20250327192000/NEC-V-ISCSI : FAILURE in 7h 57m 06s","accounts_in_message":[],"_revision_number":5},{"id":"f7fd7d373e4d3d122384b88872887c1264da4f94","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-03-27 19:28:38.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/5/20250327223426/NEC-ISCSI : SUCCESS in 5h 27m 10s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/5/20250327223426/NEC-FC : SUCCESS in 5h 54m 09s","accounts_in_message":[],"_revision_number":5},{"id":"bc0d2266d97aec5394a49c00b034320410c43756","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-27 21:08:43.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"6aab203978d44252728f22d96d7b4d54c587a2d2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-27 22:07:32.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"5fe5a59e36aeec7866d4fcbcfedc29a0af1a7634","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-28 00:57:28.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"04bb352940d13b975fa75158acf78f6c4cff013a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-28 01:07:29.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/65e0b45886e9498692f3a52d41b21db2 : SUCCESS in 1h 08m 44s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6b23769f604d4526b09488c81c9c8e62 : SUCCESS in 1h 18m 18s","accounts_in_message":[],"_revision_number":7},{"id":"8630668bf0d194bc849cceef9725544960b99596","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-03-31 14:40:05.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"8ea65e8f3e254b03fa1a83b077bbf3c574a52381","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-03-31 14:40:12.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"a8cbab4460f6a00d6f6e3aa051757fbb7cfa0f34","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-31 14:40:14.000000000","message":"Patch Set 8:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1631/","accounts_in_message":[],"_revision_number":8},{"id":"af4608a74e5747f84537c6afa49af5605dca721a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-03-31 14:40:17.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/841315/8/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/841315/8/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"ea67d50f1c121a2ae08f24d0e1a8deca0edc3a83","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-03-31 14:41:18.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver hpe-fc-driver : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"5a8cf281715db2f84c25287515df003a94ba8901","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 14:59:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5459ee5a105e473f867ce21860121179\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d07496ed77c4cc9b7fb43cf84052e21 : SUCCESS in 17m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dcde7c4485f748db81ad341034d416dc : SUCCESS in 13m 57s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1ca8dcdcca98c56584e69ee18e4f028fad4a88db","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-03-31 15:16:15.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/841315 : FAILURE in 34m 07s","accounts_in_message":[],"_revision_number":8},{"id":"5db4a61332d82f41c959c34510655d05564cb1c8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-03-31 15:18:06.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/841315 : FAILURE in 35m 47s","accounts_in_message":[],"_revision_number":8},{"id":"50af9b9448c308d744e59b99f47aae7cd0db417f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-31 15:43:23.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/528/.","accounts_in_message":[],"_revision_number":8},{"id":"d20164219f544953e20fa9b2be50750f8b3a763d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-03-31 16:07:22.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/aca04b6643384537a59b3a7582b3e2a2\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e4e3f8728bc34d37a8a2dc99a90ec50f : SUCCESS in 39m 27s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/132687a3fdc64282a3c17add4cb74c0a : SUCCESS in 45m 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7362b6e8c0e64f1293cc8973da228084 : FAILURE in 1h 26m 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b2782aeb0c5c4ec391ef3cc940d3e25c : SUCCESS in 41m 42s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/857d8b2679a841ff97d356ee9d0c09a3 : RETRY_LIMIT in 30s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6dcf1952c28e4fb8969fe1277d93473b : RETRY_LIMIT in 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b6c22d4ff648445daa39f9a9b3b24311 : RETRY_LIMIT in 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f9157435ea2e42b4b933f93d8c451792 : RETRY_LIMIT in 31s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":8},{"id":"3c235fe1ac335876964c1460b98a88b017237ebb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-03-31 16:19:54.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/612/.","accounts_in_message":[],"_revision_number":8},{"id":"222b76c5957e3a81e659c7cfb3399accb1196fbd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-03-31 16:23:26.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6d5a26f7966f452995a1636dcbf91ae4 : SUCCESS in 1h 09m 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/08ee72a053c74744a93c409c463138e3 : SUCCESS in 1h 14m 13s","accounts_in_message":[],"_revision_number":8},{"id":"99e188ca48f0b035b75eed8e61fa86c3f27d6448","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-03-31 16:55:18.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-841315-8/FC/ : FAILURE in 2h 14m 49s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-841315-8/iSCSI/ : UNSTABLE in 33s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"49aac6a04c005962a06e6020b52357c60cdfc4d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 17:17:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eebdfcfa8ede48429a04c7ffc4d596e4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b53aa5645094c30853dd6532aa7f2a1 : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f30f986dd0b84b4aac0fef13d72fe7b0 : SUCCESS in 12m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b420f733dbe4267861220e71c8d5876 : SUCCESS in 12m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddad9ca300b94fd1a398574b859e92a7 : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/9c0dfb65f7c841a796ac0452c095c1bf : SUCCESS in 28m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d05335e554c74c2d90bd4231a5b1752a : SUCCESS in 28m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a4b814d048624513bc95e597d3786908 : SUCCESS in 1h 16m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd61a4f7d12e477eb9cf39d0e315a71e : SUCCESS in 31m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/748ed4c940184628819fa4f876b5670e : SUCCESS in 16m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0cb913c56242410e93a7e9d3dd7faec4 : SUCCESS in 3m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9e1ff099cfdd426ca10dffccd9f7603f : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/371601153d9a48b99ac49b01f46ee5d8 : SUCCESS in 6m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a74b14ee862a41ef95c464f17b3cb93c : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1ba969552e4545adbf89fa5864abb1c2 : FAILURE in 35m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bfa25f9936944b589396c7677c026947 : SUCCESS in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/63d23f6794f646d8a1bf274a49d64418 : SUCCESS in 2h 09m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4de7597294ca41879b093eeebad72533 : SUCCESS in 1h 28m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e7d1adff15bb47419e65e427c5bc1c6d : SUCCESS in 2h 21m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/40ce4f1169af46daa7e02faafd4a7f7b : SUCCESS in 2h 31m 38s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3c57546b5aaf4f629b0b863ceed294a3 : SUCCESS in 31m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4db119b99c984169a6b02fc7bd57c176 : SUCCESS in 1h 04m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6259813007eb4d62a8cee02d57153efc : SUCCESS in 2h 15m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8c42e3b3794947f78dccc4e7ba4acaa4 : SUCCESS in 37m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d7bc724e3b164a99943a1aa5576459a0 : SUCCESS in 1h 22m 27s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/dc6fff6418794d3d99f4a5f0321982de : SUCCESS in 1h 24m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7a969ace22234809807ced59649fbd4d : SUCCESS in 1h 38m 08s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31f9f4122cbc4242a7bc555387e5754d : SUCCESS in 1h 18m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c348c6bcaec74b7ca9fd6931746fb61b : SUCCESS in 1h 01m 36s","accounts_in_message":[],"_revision_number":8},{"id":"4c615eb4a91443d73e75cc3c7f44aed572ce5ad3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-03-31 20:24:38.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"1a2aee15bea22dc5b9b021f9517ddaf123af54a1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-03-31 21:09:42.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"7f3a59bed3837e3947d7e1c913861955a5afc4e4","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-04-03 10:49:14.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/841315-8/PVME-iSCSI/ : SUCCESS in 1:56:47\n- PVME-FC https://os-logs.seagate.com/dell-ci/841315-8/PVME-FC/ : SUCCESS in 2:07:01","accounts_in_message":[],"_revision_number":8},{"id":"a1794f95cfed508baad816cb470c92f7dd674c81","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-04-03 16:21:42.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/841315-8/HPMSA-iSCSI/ : FAILURE in 3:03:51\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/841315-8/HPMSA-FC/ : SUCCESS in 2:07:30","accounts_in_message":[],"_revision_number":8},{"id":"0adebf2ec3a2c346d2256c27c62f3e4d79cf288d","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-04-03 20:47:44.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/841315-8/Lenovo-iSCSI/ : SUCCESS in 1:57:22\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/841315-8/Lenovo-FC/ : SUCCESS in 2:07:24","accounts_in_message":[],"_revision_number":8},{"id":"a17b98b1fe11e586c53f9271ac042abf01d342de","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-04-04 01:12:31.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/841315-8/Seagate-iSCSI/ : SUCCESS in 1:57:08\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/841315-8/Seagate-FC/ : SUCCESS in 2:06:43","accounts_in_message":[],"_revision_number":8},{"id":"9c1047a9ab37e03e93b9b556660b837b1f0a8041","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-04 14:20:43.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/8/20250404184117/NEC-V-FC : SUCCESS in 4h 29m 36s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/8/20250404184117/NEC-V-ISCSI : SUCCESS in 4h 39m 23s","accounts_in_message":[],"_revision_number":8},{"id":"2ee45706030f1c574437031be4f1187f53913b88","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-06 02:25:02.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/8/20250406052842/NEC-ISCSI : SUCCESS in 5h 31m 38s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/8/20250406052842/NEC-FC : SUCCESS in 5h 56m 17s","accounts_in_message":[],"_revision_number":8},{"id":"b355004ba4d8e3ab156e4da174d2dbb146ca8661","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-04-24 14:50:13.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"056a41631d18b63fcc9ed5f403316195775ef355","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-24 15:00:57.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/777/.","accounts_in_message":[],"_revision_number":9},{"id":"e2bb364124ba5a774ed257615bfe1dc3da70c882","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 15:24:38.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b004ded098c3472c9c3eb3e31ee2a5bf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a2f453c90cdb47d99d54431e58a88b4a : SUCCESS in 22m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/10b58f13db454c66a4c71fe7665b1098 : SUCCESS in 21m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/728e729b13f04af5b8b828ecdf08251b : SUCCESS in 18m 00s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"380e3fb80ff82517ffe2c57b8a4d2955f45c10d3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-04-24 15:38:50.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/015a1df4518a44eab18494018cda37c9\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fb174d05e2e742debd091dc1826acb56 : FAILURE in 44m 52s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0badf049435b4cd19cd14147fd360eae : SUCCESS in 42m 44s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/80db458dac2e42e1ab2f7e527193575b : DISK_FULL in 47m 41s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ca3da1b1a3fc4f918ea0b028c4a74ba7 : SUCCESS in 44m 43s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d98f659a7b81465b8748ccc63a82294c : RETRY_LIMIT in 30s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4db98d37067d4f239b4192bd362b96e2 : RETRY_LIMIT in 29s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c9bf3580e98a41809e91cfbb6271a842 : RETRY_LIMIT in 33s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cd80481163734bc4a283ce816a7ae741 : RETRY_LIMIT in 29s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#514\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#471\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#344\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#344\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#469\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2025.1}} source: openstack/devstack/.zuul.yaml@stable/2025.1#512\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-de... (truncated)","accounts_in_message":[],"_revision_number":9},{"id":"8460000049144381d235eb42d4f6a921ccc2c3b7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-04-24 15:54:41.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/841315 : FAILURE in 36m 22s","accounts_in_message":[],"_revision_number":9},{"id":"ed139bf390838a86b3fc0764236554772777b87b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-04-24 16:23:43.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/658/.","accounts_in_message":[],"_revision_number":9},{"id":"692d5d21d9ed5e6415af6f2c0b337f998b909d0b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-04-24 16:24:06.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/841315 : FAILURE in 34m 42s","accounts_in_message":[],"_revision_number":9},{"id":"e223f1e61a9aed38a765ef51020651a326c256cf","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-04-24 16:54:01.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a4f6987ec8ea4b74a9957839fb77187e : SUCCESS in 1h 16m 57s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/abed9d02b4644d3a89628d46e820de7c : SUCCESS in 1h 16m 10s","accounts_in_message":[],"_revision_number":9},{"id":"61d8dce3924cfb32aab5f63016d2445091f21c88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 17:22:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d27a78cdc524b5b8002fe728438f1a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d315d4d27b04c39b46a40099f3bdaa4 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2367646763e4123accad8a8cce06546 : SUCCESS in 9m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b04d7fc17fa14e50912e6b25d5d371f8 : SUCCESS in 10m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6882a523da1a451788985b37874458f4 : SUCCESS in 13m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fbe9668d9274a9b93add4918e0e8aa7 : SUCCESS in 14m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/c51cb21d194b45d99fee8e84ddb2506a : SUCCESS in 54m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/80bb795e3ca0493bb4eb1c842c581069 : SUCCESS in 53m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd944236d07c4ce9bf03badc818fd637 : SUCCESS in 1h 18m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca34d99bc6134e2eb6d1e25851e8befb : SUCCESS in 53m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/48554a9bfded4c99b83c6d3178408518 : SUCCESS in 24m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/19045e6067064446ac9a506e98561660 : SUCCESS in 5m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/febdeb0ab3364590895a4b9656eecace : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e06ccdfd1af04daca39ce654df1f18cb : SUCCESS in 6m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e0a2beb870b846cfbecfd3a64c854888 : SUCCESS in 6m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7b0d1107722e492887a58ba2cb1a45f8 : FAILURE in 35m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7abb918a294b4b768d567e6030a60728 : SUCCESS in 5m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d0bc5862d9974af09bb6913f3de23073 : SUCCESS in 1h 02m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/79b2714d576b47e594947d11fdc2d6a3 : SUCCESS in 2h 20m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/018f9b7bae3745c6969edbd4c033502f : SUCCESS in 1h 42m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/223ae94249a745078a3ad45da3bb774f : SUCCESS in 1h 45m 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/63c88e7751ef43119120a9646b3bc6d6 : SUCCESS in 17m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a43a2370e8854f4a89882c03bc676dd8 : SUCCESS in 1h 04m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/eef85b095f724ae3ab317c450c2bd5a0 : SUCCESS in 1h 36m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d64cce70eac940db8c4f8523e5f883e1 : SUCCESS in 35m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b5b5d0f6a7d94c28af09a5b68b3c012a : SUCCESS in 1h 30m 34s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a45d16cadae444439eb2a8463e7d5618 : SUCCESS in 1h 44m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3ee83d5bee6040d9a88bab5a7f2f2993 : SUCCESS in 1h 11m 58s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27169e4453d348708d9e480c6fa9692e : SUCCESS in 1h 37m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc1361f8bfce4255a1b6d2fc6f0d8f21 : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":9},{"id":"9863af0685d605e9bfc9261d27296b1364e6bead","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-04-24 20:29:18.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"e4bf93910ff9ddffaebee8ba56fbff5a2d8ff361","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-04-25 19:04:53.000000000","message":"Patch Set 9:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/9/20250425191141/NEC-V-ISCSI : SUCCESS in 4h 08m 43s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/9/20250425191141/NEC-V-FC : FAILURE in 8h 53m 09s","accounts_in_message":[],"_revision_number":9},{"id":"f21f1e36a39d2c455bbaea10c5f87615b090944f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-04-26 21:56:03.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/9/20250427005950/NEC-ISCSI : SUCCESS in 5h 26m 33s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/9/20250427005950/NEC-FC : SUCCESS in 5h 56m 08s","accounts_in_message":[],"_revision_number":9},{"id":"0250f9cdba1bcbe9a48e36d5f96c40e6e5973331","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2025-05-03 15:03:48.000000000","message":"Patch Set 9:\n\nBuild failed.","accounts_in_message":[],"_revision_number":9},{"id":"b4349de66916c3e103b07a06b3671b7dd8ae33ba","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2025-05-03 15:10:43.000000000","message":"Patch Set 9:\n\nBuild failed.","accounts_in_message":[],"_revision_number":9},{"id":"50f9df99d01a75251051c03d809c19f5d522baf5","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2025-05-03 15:17:38.000000000","message":"Patch Set 9:\n\nBuild failed.","accounts_in_message":[],"_revision_number":9},{"id":"7c6cf90c805c9ec03a8aa089dd55cb86c45ffd88","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2025-05-03 15:24:54.000000000","message":"Patch Set 9:\n\nBuild failed.","accounts_in_message":[],"_revision_number":9},{"id":"a6c2eacd3223fc3a0c71b633c7e600bbaf1b5bb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-06-26 11:38:03.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"ebc77db24abd37e9a2c9f9fd52055db48ce47aa1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-06-26 11:39:13.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/dfaffa879732482297caee560a446b4a : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e160f057a93c4e6899b708a80be86fc4 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"38b02b862c758e04d8a1de015339e60da0c816b4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-06-26 12:34:16.000000000","message":"Patch Set 10:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9c09e1016e224ce2acbbf0d9abb70d9a\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8499011fac7641d48969366e9963f18c : NODE_FAILURE Node request 200-0000069465 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a9387fd3de4e4a91953d080c687bd305 : NODE_FAILURE Node request 200-0000069466 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e2fc5fe4fa04476b89d29b8b2f572048 : SUCCESS in 43m 36s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fafe2369d4bc459a98c089963e19111a : SUCCESS in 45m 59s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d37386a3094e4d93988670226699a06f : RETRY_LIMIT in 31s","accounts_in_message":[],"_revision_number":10},{"id":"16ee4d08c999e1c0fa7dab85c24238b086d572df","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-06-26 13:20:51.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"934c1538200f031bedda9f17761f70d190df373c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 13:25:59.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71496f9f8e414e5d8f23522eeb3347c9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e2cdf0b1209145228c0cceabe61fe7dc : SUCCESS in 23m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aa52335abd6345819b6ce00ff34c17a1 : SUCCESS in 21m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/da4b0096d4254fcfab146de1653b7a47 : SUCCESS in 28m 10s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1619efcf7f6db6e001e6ea6fd1108456400bd47a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-26 13:34:38.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1208/.","accounts_in_message":[],"_revision_number":10},{"id":"2549d3af76d89556332f4a3629491e20b645d5e8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-06-26 13:35:41.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1079/.","accounts_in_message":[],"_revision_number":10},{"id":"43acba96b6945576976212b4c0d5ef2a2e05c8e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 14:41:54.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3680b482d1bc44b58bdadc0c63d7e2d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5992461d490d4d46a782d661637af971 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aace06e029df4f8d8c69c68be2b7f760 : SUCCESS in 10m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5fd1e79777f242c99156055680c045d5 : SUCCESS in 10m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/491c4751715b48ef81b982d6acad714d : SUCCESS in 13m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f18444c8e7824300947ac364d1a696b8 : SUCCESS in 7m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/d7b50384fe264c139991be4bc12dc278 : SUCCESS in 57m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1044d3c19c134b3a8fde08d815123441 : TIMED_OUT in 3h 02m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2179ac7290fc4a2894a761a7a5e8926c : SUCCESS in 1h 22m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/452e7949b1be445ba36d3a197f045a22 : SUCCESS in 48m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/052a47ce97b24f73ab6dde5eb68f5855 : SUCCESS in 21m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5d84388bddfd4546aeb00a60754acbc5 : SUCCESS in 5m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c461d3997c1945d59fd1299ab24b5e72 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f15c0b9596a94733925a3863d4cef458 : SUCCESS in 5m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/84a997f77ddc4325b1b21e42b265a7f8 : SUCCESS in 8m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2125e74716b346f18cd5a50d336971ef : FAILURE in 36m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8631788966484c71afc6e1ebe8b1dbb5 : SUCCESS in 7m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ad3b8749d1ae4126aa9c182b8b39b269 : SUCCESS in 2h 00m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/742857307fd9459488d761bb3ac9abb1 : FAILURE in 1h 30m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/13a3bf15597a498e91eb24342c074cde : SUCCESS in 1h 42m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2e6ddf86aa304c408e996ebe6a040bce : FAILURE in 13m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8cbfb2a8225b4a5e941e31e59931028e : SUCCESS in 17m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/43fda5b7af2f4df0ae08228bd93990af : SUCCESS in 58m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c1c5ff5984fa44f787dabf5f92a95282 : SUCCESS in 1h 43m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2b7c2784a1b24a5f869a41f309b02210 : SUCCESS in 44m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/47966191d8e44cf08fe5d6f573f37041 : SUCCESS in 1h 16m 22s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a9ceba3eeae54b8e9f8c8db61fb38d08 : FAILURE in 17m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b51a6a63b08943ebafe1c1ab13fb3b94 : SUCCESS in 1h 26m 31s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c7e97fa0580479c8e2baef4c808cdfa : FAILURE in 1h 38m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c802f199ba2c4e4d94ab86624a711283 : SUCCESS in 56m 08s","accounts_in_message":[],"_revision_number":10},{"id":"3ba462e348934edba8fa475750c0163704a49575","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-07-01 16:41:02.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"99a78a412c30b7bb53a6238b7adcf6922d6493fa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-01 16:42:07.000000000","message":"Patch Set 11:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":11},{"id":"6f03aafcbce59393a4d33a4c54e0370b189b29d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 17:14:08.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbedb81738634089b4a0a06b4fa3f31f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ff5d843b13714402aa5341bccf3ed38f : SUCCESS in 18m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe6d8296468f497088bdd768e336d8e8 : SUCCESS in 14m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/01946528b6994796a7d8eef6a703ce3c : SUCCESS in 27m 22s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3d109233f867aa33ea80cb427b348afe44270cef","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-01 17:34:56.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/bc7f668b529e4492bab306d4c1e32d6f\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c470421cdcbc44cf9d5f3bf090fad8ee : NODE_FAILURE Node request 200-0000070016 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f5b36595bc294be086fbfa3ca9982900 : NODE_FAILURE Node request 200-0000070017 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/01ef1e4a4d5b44c1bcb108f32b07cc67 : SUCCESS in 53m 07s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1652ef6394064ffb9962346bbc1761ab : SUCCESS in 47m 23s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/25f60b15fd9a42f699f69819142b549e : SUCCESS in 46m 05s","accounts_in_message":[],"_revision_number":11},{"id":"497a89b98e0d568680a37a08cc52e93f62d99e29","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-07-01 18:23:46.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n\\* cinder_quobyteci (NON VOTING! This CI is currently under construction) : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"ca3089a825834d4c1b3e4576d5e1a364f31b1117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 18:28:52.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/06064a4708ce4dfd878466a63e0cf832\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e2e0b1d2ff84c49adfc3631718517b4 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3285bd8b4dfb460ea9c8ed51a2fa850d : SUCCESS in 11m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3fdac441014b49eaa689fd531a3799fb : SUCCESS in 10m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/893f86b2e03b4d77a2329957800795d9 : SUCCESS in 11m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8b79a4b13914c60842494938882beda : SUCCESS in 13m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/6071600ca9e44aa097007aac42a870b2 : SUCCESS in 59m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/85742b5842104e43a81774b9c2e26121 : SUCCESS in 1h 01m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dc3d4ce5d77d45f9b34a3efdc3fab681 : SUCCESS in 1h 13m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48e2fdd095b14abeaa007098c0a9056c : SUCCESS in 1h 04m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/197ff65f853b47ce83474009468075b4 : TIMED_OUT in 40m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/951c1fa5971a43fc9740701a0e7147db : SUCCESS in 4m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/999253dd8c044fe283767b49c861b7e5 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b1830e02bb2d44f38fc78d1a38d2b1c0 : SUCCESS in 5m 23s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6a06eabf99a042b994d12954a17f7571 : SUCCESS in 10m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fb98ce68342e4ec2bd99d859f1c3e6a5 : FAILURE in 31m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8eccbe9751364f6d9e73dae520b9d1b9 : SUCCESS in 5m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cc0a983a725b48af87757ac19c2f6b91 : FAILURE in 4m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5dff84b0facc4627a38b5fb95f51e4d2 : FAILURE in 8m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a65fb5a85b35457aaa93b15eeac16a04 : SUCCESS in 1h 46m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/47a806dcdaa64351b4a192501d9e8b23 : FAILURE in 20m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e85fb8186b3144c6a64cfbea4a58ef33 : SUCCESS in 26m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a4cb167a962b4de0b5edb6e6c947f685 : SUCCESS in 1h 16m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cbf277b812e34938b98b757ab6d2d498 : SUCCESS in 1h 34m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4e1bf4d6018340abaff5a3c6bd891030 : SUCCESS in 29m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cb0260519c004edeafc4795ba6eb238f : POST_FAILURE in 1h 29m 38s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d43ccabae8a947a48047d675fdaa8887 : FAILURE in 12m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5751b2f33f984fd3be2c54f95867a2e5 : SUCCESS in 1h 26m 06s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f46dd4849ae94e3a85415cd3b75e30da : SUCCESS in 1h 28m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb836299c941458dab67ca8798370f6b : SUCCESS in 55m 11s","accounts_in_message":[],"_revision_number":11},{"id":"55634020b19bb6a564d714cc9c707d95370da96f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-01 18:31:50.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1254/.","accounts_in_message":[],"_revision_number":11},{"id":"ad5c7aa0267a947194d83762113addc03855ea9f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-01 20:24:58.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1125/.","accounts_in_message":[],"_revision_number":11},{"id":"f32e1a87469d4701107b2dbd325c8580c07350db","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-07-08 01:19:43.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/11/20250708011800/NEC-FC : SUCCESS in 6h 17m 21s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/11/20250708011800/NEC-ISCSI : SUCCESS in 9h 01m 40s","accounts_in_message":[],"_revision_number":11},{"id":"1f77ed4bd35cc0a82bd9fa36dca2b1cc87547ad8","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-11 18:26:27.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/11/20250711215926/NEC-V-FC : SUCCESS in 5h 19m 22s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/11/20250711215926/NEC-V-ISCSI : SUCCESS in 5h 26m 58s","accounts_in_message":[],"_revision_number":11},{"id":"d70eb2212d8cedcc3ac2f1452d6f4a1e7ab152a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-07-23 12:43:39.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"c66b4807903a88e4be124fb9ac8c6ebefeca52d3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-07-23 13:21:35.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/841315 : FAILURE in 35m 44s","accounts_in_message":[],"_revision_number":12},{"id":"f83eaa7501db2b11a3efd3f9340a065ab866426f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-07-23 13:21:51.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/841315 : FAILURE in 36m 18s","accounts_in_message":[],"_revision_number":12},{"id":"6d4f02231130acc25cc92858c8cfe04c2b12ac61","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-07-23 13:35:52.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3bc8bd5f55f04759806e8a319a78895d : SUCCESS in 44m 37s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6d3136fcdcfc4c7b9c241f0b97025422 : SUCCESS in 50m 11s","accounts_in_message":[],"_revision_number":12},{"id":"3fa7b98f75aa048304594869b1b0ad838f7735a4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-23 13:46:41.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1300/.","accounts_in_message":[],"_revision_number":12},{"id":"4a377d2b4be66eca3fc99df6bfbfafa6c54bebe6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-23 13:50:01.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1433/.","accounts_in_message":[],"_revision_number":12},{"id":"8471438b7dcdff48da724dc22a88be79d0276f19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 13:54:02.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2aab9e7aa00d41f29c3187db042e9b44\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fb6d659236664e5b9a304dbe39da5544 : SUCCESS in 25m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/250dff1f685e42508d2d002a8230b3a1 : SUCCESS in 25m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/46fa79749c744af9baba0341860653e9 : SUCCESS in 28m 36s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d5a53a8163bcb89ff71e88655cfbe2b8ec58f5bf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-23 14:20:32.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/facf23d646cf4d78870073d727c886be\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a0ea984aef34461ebe0053718c39ec3c : NODE_FAILURE Node request 200-0000076215 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/75659d2da9fa45dbb98e0d80f6fe2d27 : NODE_FAILURE Node request 200-0000076216 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8650549fc67b4e9cb4b8a304c913328e : SUCCESS in 58m 35s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a501dadbff944f4595082fa2aa63b402 : FAILURE in 1h 36m 02s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/52cc10575ebd4ab08fde51f127c74ddf : SUCCESS in 1h 04m 43s","accounts_in_message":[],"_revision_number":12},{"id":"e4db6984acceda98d23abff4080dc20e601a2f3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 15:30:10.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e43b0f050db24158913b96e2894d37e6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45e89f93c8f24254b13bd33526e45ecf : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f8d22cc546545c098723471fa6249e0 : SUCCESS in 7m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/62ba9f66808a40d69f90be981849c230 : SUCCESS in 11m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7181fa23545a47e1bc31e005dfd6b501 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95f22aa808964059acd3ba052f789d3c : SUCCESS in 13m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/4241f117bad34155a116728cf055df1e : SUCCESS in 29m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/34a4087d54a74804af7b88553ad05817 : SUCCESS in 52m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ca68feaafa30443d9bf517829a4c33ec : SUCCESS in 1h 41m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ae256ccb984445f8a04a918a0888981 : SUCCESS in 1h 02m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e2cf0a357051444a9b0dc937b15e1c36 : SUCCESS in 16m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8a64df058f014f4981f387b8414d59ef : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b1c945dea72a44c9af854efc06d26a8b : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7206534da0e34ef2af110a16b0550510 : SUCCESS in 5m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f1dd946998d04bc2871c85b4cf873f51 : SUCCESS in 8m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0226fb0612e64b438144fb72b3d708de : FAILURE in 34m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bf42f487d5b74cf1b336ef4441f3b96a : SUCCESS in 3m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/05e47dc3a09b4b60bf76db74f0aa5a76 : SUCCESS in 2h 06m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6d5945db90d94501957e43d4f37be1ec : FAILURE in 2h 40m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c633bfa2bc4649e0b4c4b37ddc067fa5 : SUCCESS in 1h 48m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c4cfbfa813714c6dbdac6701d8a6a3a2 : FAILURE in 18m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/290ca3f87fb54d15abcd71bf66528070 : SUCCESS in 45m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4ba5293fc35643948338209c0967e8b3 : SUCCESS in 1h 07m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e07b4635895948debc9e35d3ac54f736 : SUCCESS in 1h 31m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6f972e58449948d49e2e3dcb86413410 : SUCCESS in 34m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2af21000310f4b8ab21dd24acaef7857 : SUCCESS in 1h 25m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/dc5fc4b58d5c4cc8b71d5f9fe45df6bc : FAILURE in 22m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e83f434ed5fa4b80950d55705b403097 : SUCCESS in 2h 02m 09s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5675ed9b13b4661a97da16ac06c8989 : SUCCESS in 1h 37m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9559c3e1596d40c1bea1d7f41a1c40be : SUCCESS in 1h 03m 44s","accounts_in_message":[],"_revision_number":12},{"id":"839aed11d99149e95f8cba529f27df9fa3dc1e50","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-07-23 15:35:52.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_07_2025_15_30_23\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":12},{"id":"2a4ace6853290b9baa7bb012d64d2e5a42446897","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-07-23 15:38:22.000000000","message":"Patch Set 12:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-841315-12/FC/ : FAILURE in 2h 46m 1s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-841315-12/iSCSI/ : FAILURE in 2h 54m 21s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"b518c475801935be9d10f20b09252bd46f545a0f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-24 14:17:09.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"8451c41ab9bb09fc330e2958b73819c5f9cec86c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-07-29 05:41:50.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/12/20250729085306/NEC-V-ISCSI : SUCCESS in 5h 20m 22s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/12/20250729085306/NEC-V-FC : SUCCESS in 5h 48m 42s","accounts_in_message":[],"_revision_number":12},{"id":"dd8f4edc554e52f86cba3a0e297c8fae65bb8bee","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-07-29 19:58:19.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/12/20250729224133/NEC-ISCSI : SUCCESS in 5h 50m 28s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/841315/12/20250729224133/NEC-FC : SUCCESS in 6h 16m 43s","accounts_in_message":[],"_revision_number":12},{"id":"f8050736d3a5e151bdb54e79f1c917866fc32341","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-09-29 18:35:09.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"074935c985cab1da9e002570487217b545032aef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 18:56:48.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45985b41f85f43aaa3b6d74c45b52db7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a530e49f12e04063957833e57cbacc18 : SUCCESS in 19m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fdb0f23cf1ea4692a4bc32a7ecf6b2b7 : SUCCESS in 19m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1878d7a0946a43178262c2e0f0fcc1aa : FAILURE in 16m 14s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"91de89c806923c01b4fc171300d97d9ea2058bfd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-09-29 19:14:02.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/841315 : FAILURE in 36m 50s","accounts_in_message":[],"_revision_number":13},{"id":"9f29bc210c4430711609d6339aea7979c69b255f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-09-29 19:14:43.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/841315 : FAILURE in 37m 47s","accounts_in_message":[],"_revision_number":13},{"id":"8b6bfdbbf72cd31638cecdfbe762da9bbb6a817d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-29 19:49:07.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2426e9373ab34ae886c59b33133bbb7e : SUCCESS in 51m 00s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8ce50c41b3f84a76aebfcbae2571818d : FAILURE in 1h 10m 14s","accounts_in_message":[],"_revision_number":13},{"id":"a0b8dcf26e0b7100ffb98acfcfef87ff6313df78","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-09-29 20:27:50.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ide947c8e3d3f050665c3acb804b6617d2b9c045c .","accounts_in_message":[],"_revision_number":13},{"id":"c04fa8142fe30a576a0cbd0a54b7ca8338fc49c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 20:43:05.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/021901094ed24c5fb954487ccece1204\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b3dfbd0183a486bbdd5931c28d9186b : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf1ae65bc5254cf2bd44a2bf478d0e08 : SUCCESS in 9m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1861bba70a034199a9c1aff0abc6f841 : FAILURE in 8m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b2fc26a19d24328bf0dd2853684efe3 : SUCCESS in 14m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aab2f84c84643f5af8d1965cd104c14 : SUCCESS in 14m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/aa5193b155c94ade82284a4f80de8e48 : SUCCESS in 1h 01m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aedd33c8c31c4c66883cb2694fdb0e72 : SUCCESS in 1h 01m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/02c774dd6d4049edb28ec4991161dadb : SUCCESS in 1h 35m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/116383d457bd42449595240dcff9d1ef : SUCCESS in 1h 00m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0f6b973c18c94fe7befebf107a592bdc : TIMED_OUT in 40m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a1df52c06a5a412e9020ef95e48a4c9b : SUCCESS in 6m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8b0f41f137944d7a97902533d9fb964f : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0898cc250b8f4dc08293b3b1364c70f6 : SUCCESS in 7m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b4b112ff308b448082f135c88780369e : SUCCESS in 10m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/19fba78a20df47ea9518c66753611466 : FAILURE in 39m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/39e40b86e7234824b9f02d3ca199bf19 : SUCCESS in 3m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3791cc7059614c16bcc7c75bfe1cbdc9 : SUCCESS in 1h 59m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a967339444924abc86eb9136d19e9937 : FAILURE in 1h 39m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8e5de6c42a0c4ceb9e4d7a0b5d424a99 : SUCCESS in 2h 05m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/9cda3e9a56344eaba6992e540f244a98 : FAILURE in 18m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/04b9f04db6f74f5aa7134b61a03486f7 : SUCCESS in 37m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3df80a7ed2d14a34904becb54de4c47f : SUCCESS in 59m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1636a2da61e64de893a03f73b3d14c1d : SUCCESS in 1h 37m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dc13d12364154bffa7a320701d074158 : SUCCESS in 27m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7dd698688d1240d2b540f94714e03b13 : SUCCESS in 1h 18m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/526081f3805d42f58be45f820cc94ca9 : FAILURE in 18m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/24b9213814cd4060bf5da2c217d5fe39 : SUCCESS in 1h 32m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42cfabd9885f4077adc6c621442acbc3 : SUCCESS in 45m 57s","accounts_in_message":[],"_revision_number":13},{"id":"0ec4d6f2fa0de9836b9b6205ed3b00e245907203","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-09-30 20:01:03.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"27a6f8b48893f0918e081be07bec0ce5c8d28aef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 20:22:04.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bb347e7cdba4cbdacbc2743d6772fa3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/64336fc9b3e24ad0a85a7c0156f351bd : SUCCESS in 13m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b3838efb43444d7f9fc87daef519c1e1 : SUCCESS in 17m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/edf46268f77544439bf8fd25feee2ab8 : SUCCESS in 18m 54s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"61859eb3d4efdade41a20982f90d961de69d9db6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-09-30 20:59:57.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c53bf9e1b1374e7fb3d3672e919abb17 : SUCCESS in 46m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/051a1fe99f5c4bff9da2baf8c1980f61 : FAILURE in 55m 12s","accounts_in_message":[],"_revision_number":13},{"id":"028b54d6ee5b2eea52c98e61094430a8474c8cff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 22:40:36.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc0d3fd5034243ff9a80a3efc7db6e6e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b59ffc8862e403a953e5a4bb297d18a : SUCCESS in 7m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa6bb0c809ed4259835c1cb03aa34d6e : SUCCESS in 12m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c5cf45f6c34c4081916c7736baa7e012 : SUCCESS in 10m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9fa762dcf797489e9c918fcc4411fc07 : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e875f6f80e2b4bf78ee1cb6466d597d2 : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/eced8ec24359489ba4196420d9c2df1a : SUCCESS in 58m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/703f55c77a434e93aee0da7a6801751f : SUCCESS in 58m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9e779ba1352442bd976fc21018f30668 : SUCCESS in 1h 23m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/820d9587d11344c99dbf6771e8cf5831 : SUCCESS in 59m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9c03d2402a1c4e73a9e70efd7032918e : SUCCESS in 23m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ac2eddff5d734a6f98d1f00a9f1b08e5 : SUCCESS in 6m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0c92d964ad9d4e64bb8ffb3f2486b186 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7819cc2955e84501bfb1339b5a489942 : SUCCESS in 7m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ff3bd1d136cf407490192639be6f3c6d : SUCCESS in 9m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/361c7e4e053f4dc9937a4a8f0b73b7f6 : FAILURE in 23m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/38437635457e44909772f74d8c333728 : SUCCESS in 7m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ea3a1f2b0a3642fdbf0325b111194442 : SUCCESS in 1h 51m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/174f789fc3d64cd28e8c7c56fe53020b : FAILURE in 2h 32m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5506e6db02b2412499160af603238ad7 : SUCCESS in 1h 20m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ee73fa067a144732aeb3f3150d142e12 : FAILURE in 15m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ad97fa2796a44f2985841f1c6e7a1acc : SUCCESS in 36m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d163649d7744455290aed0e7b95c1859 : SUCCESS in 42m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/75e6e4bbed984f728722360f94bbfc12 : SUCCESS in 1h 04m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/815aa325d74b4b1887dbe00f010a4ac3 : SUCCESS in 44m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3127c58180eb42bcae22704d41b43dab : SUCCESS in 49m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f665114d60e44f44a177a39f37c47953 : FAILURE in 10m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7ab474e608b0419199b86f3a351e86ed : SUCCESS in 1h 36m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f59fa5775dec405faf2e706ec64d2683 : SUCCESS in 26m 47s","accounts_in_message":[],"_revision_number":13},{"id":"5b4b262087f4a3ff04e1f985b48802baf7be86bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-09 15:10:22.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"d22c292f87ec57b3764e230ac59fd0b02ab01ee8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-09 15:49:04.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ide947c8e3d3f050665c3acb804b6617d2b9c045c .","accounts_in_message":[],"_revision_number":14},{"id":"b9542c990a3bc582d9c156311c2dc4def9229c37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 17:28:37.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62e0687d28454e49bf752294e12ea932\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/949936a94f7e4bd5ab75f363dd5bb806 : SUCCESS in 24m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/80fab86add984f4cb0f36d1b2f936bad : SUCCESS in 27m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8a3cf0aa009745539cea909a1cd59dfc : SUCCESS in 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"ab38c8e2ee60d9862d257203077c3e8bdf1ddf5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 17:38:58.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57f798700629413aad99b8aea1e51892\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5af49dbc4fd84183b74bd795f249e6d3 : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc8ba35b63b74fa3bff4f3be14c0dec5 : SUCCESS in 12m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/178f095485494708b0a7e1e5f82fcd5e : SUCCESS in 10m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/adb19b59a008400f83e758e1fc80f44d : SUCCESS in 13m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bdeef8e09f141a89e917ec58bdfc2b9 : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/7bb377665c23445c98b70de0be98b0b8 : SUCCESS in 56m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eee70736cbd545ac9684db2c594ca1f6 : SUCCESS in 49m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2572fba621ea4c69be16ea7b9c685e54 : SUCCESS in 1h 24m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d75ba1b416640c29e4da8122db82f24 : SUCCESS in 1h 04m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/375b2590205c461eb2f0efccd939f11c : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cbb0c21bb4a74c56837be57cf3ee42fb : SUCCESS in 3m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/86a7393a03f94ac3895f2d9bf16647d5 : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7d87a5575ede4d02939802cd12319d18 : SUCCESS in 8m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ddd31f9f116c4101b4a9fb8f30eb9441 : SUCCESS in 11m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e7aeed64da93496089d1c9668df0de8e : FAILURE in 37m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/467809233e3a4de183e35c4bd5a04af8 : SUCCESS in 6m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/02804f405dae40ac9b791987c8334baf : SUCCESS in 1h 40m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aee43f54fba143569c928310fe6e1056 : FAILURE in 2h 23m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4f6aa4da078b4bea8198a086f6e9c121 : SUCCESS in 1h 24m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/684dc586216d4d8198f944cc2aa540ee : SUCCESS in 1h 50m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c4ad8b1ff36f43dea1a24e7ae3a1214e : SUCCESS in 25m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7b1a97d749ca4a3fabf69529e6b38574 : SUCCESS in 52m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ea297385c68c4f2d8e2f8f9704ebd757 : SUCCESS in 1h 20m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fad5663be4e54ef9a95d27cd4fe96891 : SUCCESS in 35m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/18ea7eea83a34a4b9daa019933f157cd : SUCCESS in 56m 20s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/a803bbea57df49d3a44e31c4a09b5489 : SUCCESS in 1h 42m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b87c1bef269b4e628be73eea61529785 : SUCCESS in 1h 34m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76fec1e7983c43529a5189d43539da05 : SUCCESS in 53m 59s","accounts_in_message":[],"_revision_number":14},{"id":"f59b6cf01c5f69e8c40173084e6b57ae1fb06801","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-09 17:45:27.000000000","message":"Patch Set 14:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/09_10_2025_17_39_09\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":14},{"id":"ecfe8cc7a6784bf8e40748c3cc566bd9d673681b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-09 19:28:38.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/841315 : FAILURE in 35m 52s","accounts_in_message":[],"_revision_number":14},{"id":"a6fac1bf2a24b0401004b5316a6e492d92350e57","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-09 19:30:49.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/841315 : FAILURE in 35m 33s","accounts_in_message":[],"_revision_number":14},{"id":"ef44b747f030d8913f834bda0ac63fc6b13c6b72","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-09 19:42:50.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0b761183c9ab44c6a7433ee74507b09c : SUCCESS in 51m 21s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8bce8304c5134eadb1cd7f82c1826f57 : FAILURE in 53m 29s","accounts_in_message":[],"_revision_number":14},{"id":"c14198d372dbae9ff2aa22772b18123353489295","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2025-10-09 19:55:19.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"6ec6cf299e5467e85c64bb99768849a0aef4b64a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-10 08:36:35.000000000","message":"Patch Set 14:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1975/.","accounts_in_message":[],"_revision_number":14},{"id":"6b134ed36200ae82b8e763b65611348d48733451","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-10 08:50:03.000000000","message":"Patch Set 14:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2089/.","accounts_in_message":[],"_revision_number":14},{"id":"24ef6fc12b5b03acf12001dfdedee2726660b18e","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-10-10 11:54:08.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"1dbb338e920c00c4d1011c02465fbf528dc4ac51","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-10 12:53:26.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ide947c8e3d3f050665c3acb804b6617d2b9c045c .","accounts_in_message":[],"_revision_number":14},{"id":"2eaa0a70352389ad96012844ed8ba36e276bea66","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-10-20 16:30:21.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"53e2edeb44344d102139a4bd8436c1616713c724","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2025-10-28 14:48:43.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"72da9f6d70de8f7209adc8c3b618dc7af639d404","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-11-11 00:37:21.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"6c7c5b693338e0af9f98befe74f98f53ca22c58e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-11-25 14:43:50.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"0849659f08f0fb26ea760a2bb3c81fffc8916e38","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-25 14:44:00.000000000","message":"Patch Set 15:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/68/ : FAILURE","accounts_in_message":[],"_revision_number":15},{"id":"1651d8dc826dc62edb6e1e923d45858b873e20e5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-25 15:25:26.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/841315 : FAILURE in 36m 48s","accounts_in_message":[],"_revision_number":15},{"id":"2cb5fc58d614050c963943807a8bd18037575943","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-25 15:36:41.000000000","message":"Patch Set 15:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2574/.","accounts_in_message":[],"_revision_number":15},{"id":"eb998f59a07dd1b14ea779b9aa68a2773e5fbdb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 15:37:59.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af1d56df7bd64387b4b5ce0abab0efae\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c6d151cb946641a0896784553dc22fe3 : SUCCESS in 13m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c0fabf1bdb324e6482a3e41a1a588d41 : FAILURE in 42m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ed8155bd461144f39fe4356a53d7dfcb : SUCCESS in 18m 45s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"2e81bc4360ff3d6d88df6f46676d1c18bc8f6c80","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-25 15:42:25.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ide947c8e3d3f050665c3acb804b6617d2b9c045c .","accounts_in_message":[],"_revision_number":15},{"id":"7c2e63a8960109c5c0fac0fc48f99a6bf76888fc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-25 15:44:43.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b9343c32a20f4cb9a8b4768b4ad83ff5 : SUCCESS in 56m 35s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c5b9c73014f946ee9d6ec4cc601e998c : SUCCESS in 56m 21s","accounts_in_message":[],"_revision_number":15},{"id":"d0d6b72678d1237d7c8e9a1e903366d559fd62c4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-25 15:46:28.000000000","message":"Patch Set 15:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2722/.","accounts_in_message":[],"_revision_number":15},{"id":"9dbfc7ff0c48a2b98098e8c867a2b46128d95430","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 17:47:57.000000000","message":"Patch Set 15: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aee5044dab63425c93ad031ffe592a7a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87fa75ccc9a648fb92bd27db906516b9 : FAILURE in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5fea882a99644237add40c08484352f2 : SUCCESS in 11m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a4ecd962c28436080ccbc03ffc0cb45 : SUCCESS in 18m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/249f631ced224c44a2b229e92166fceb : SUCCESS in 13m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/3f8b5d453d5c42cf98eeaa319e8af98f : SUCCESS in 57m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4905ae3c95fa4ec1bbe96026c34da709 : SUCCESS in 1h 02m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f7fab66aff54a0bb7393d51dc847677 : SUCCESS in 1h 27m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a36df63d7f634602906e639d81baeefd : SUCCESS in 1h 01m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7db452d5bcdd4a67ac697832973f131c : FAILURE in 29m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9e5642a0c3ba4fa18aa291598079337d : FAILURE in 5m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0953665f4da94d3a98c7e2d711c7aa10 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/863b3f4457c84fefab5b2ca542c70afb : SUCCESS in 6m 37s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1f16140affb24fdb966d0330f30eb563 : SUCCESS in 5m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/938c872cb4a74d93bef95758c905d9a0 : FAILURE in 20m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ca7b897b67a94516bdf3dcf1fbf92ec3 : FAILURE in 6m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6df1260ce1fc495fa3c018b47ea5b3d9 : SUCCESS in 1h 43m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d5c8d6c62a7e4342b4807c0bce1fe3df : FAILURE in 2h 20m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/529f92bd94934e0f99ec88158700c8b5 : SUCCESS in 1h 44m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8f554c24771d4227aba77a73cb8e032d : SUCCESS in 2h 42m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c92d38a8d8eb44e29ad1cef24e79f54e : SUCCESS in 36m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/241d6811430b4db8b440a2a7f8fab887 : SUCCESS in 43m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4e059ca3e11e45488821c031ca47a9aa : SUCCESS in 1h 53m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3e58cf7e2a55436da813e25c463b25bb : SUCCESS in 43m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/32990dca801a44f0a40715978e797fb1 : SUCCESS in 1h 20m 57s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ac0fbaef735c4215a3c328e849e94a4d : FAILURE in 1h 50m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/850ffed278994f9e93f5a622ca6f31a5 : SUCCESS in 1h 55m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/522227f69da644438233aa7ba92efd7c : SUCCESS in 54m 43s","accounts_in_message":[],"_revision_number":15},{"id":"1bf83fb7016b991c4c85bd494e542bb9d0a9eeca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-12-19 14:42:44.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"4bf3e41483c407498a65c31e564702cf4fdbeb80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-12-19 14:47:54.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":17},{"id":"84117cb432010a2e9b07f27678be70615c11baba","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-19 15:26:56.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/841315 : FAILURE in 34m 22s","accounts_in_message":[],"_revision_number":17},{"id":"575f68bebed6e29ddb8c2250d6fcc51572b53d58","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-19 15:41:12.000000000","message":"Patch Set 16:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ide947c8e3d3f050665c3acb804b6617d2b9c045c .","accounts_in_message":[],"_revision_number":16},{"id":"0e74831980c0e784f97c15aa4016f61f1f264f45","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-19 15:43:28.000000000","message":"Patch Set 16:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2937/.","accounts_in_message":[],"_revision_number":16},{"id":"88080e94d89264059fccd60f2181e20700d0d687","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-19 15:44:54.000000000","message":"Patch Set 16:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2766/.","accounts_in_message":[],"_revision_number":16},{"id":"4b5682f9bda1e4d4e1069b7c1e4b429c9555a098","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-19 15:45:39.000000000","message":"Patch Set 17:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ide947c8e3d3f050665c3acb804b6617d2b9c045c .","accounts_in_message":[],"_revision_number":17},{"id":"b4a459af7e56376132a14eea184d9c28a886c1be","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2025-12-19 15:49:28.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"5ace3f9361d3b107330487dc72de7f0993b40c65","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-19 16:37:48.000000000","message":"Patch Set 17:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2938/.","accounts_in_message":[],"_revision_number":17},{"id":"a26d86f74b5c9a9868b10c792b3ede7f3d42db2f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-19 16:45:31.000000000","message":"Patch Set 17:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2767/.","accounts_in_message":[],"_revision_number":17},{"id":"d7333372e0d7704be302f40584f9927e15f7e714","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 16:49:45.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/072b5312c8e24897b7d16551dcafa4be\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/92eb49438fe74bafb44d4d51855edd0f : SUCCESS in 12m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3c06591bdb8440b79b7ad0b327c70ed8 : SUCCESS in 23m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1155e8e4ec2841f6b6de9209f2870214 : SUCCESS in 14m 54s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"ba13129a49195688251dd26ace896e3ba3a55dd5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-19 16:53:22.000000000","message":"Patch Set 17:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ide947c8e3d3f050665c3acb804b6617d2b9c045c .","accounts_in_message":[],"_revision_number":17},{"id":"e0f693dfd0ff50a3ba55c867a6400f2d7392e301","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 16:57:14.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59d31e4e3dca472fb9c7a7bd2df9dc9f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d4dd02fc1574155b878ce7ca429a4d9 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/794752b0ddb54827b24f58ebffd2725e : SUCCESS in 7m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/073b287a136e4b6590b5d9d2c898d09d : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c884ad350c724e5cb66092ac954169fb : SUCCESS in 6m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/3e4d040e7d3b41c0933608eb715feac2 : SUCCESS in 1h 00m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6b85462bae814e89854ebe8e9986698c : SUCCESS in 1h 05m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5f45f8891e54488d976834c88588f14c : SUCCESS in 1h 24m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5e8b3a140ff4e5d826b8f77e504f285 : SUCCESS in 33m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/848dbf63fb4643c7916d5c950ffd2b58 : SUCCESS in 16m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e138461330644ad8aafbb8cc13366f3a : SUCCESS in 2m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5f36d55ea8bf41e88d9347f2c436f473 : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1ad5494733e14c9f87360a870206a8ba : SUCCESS in 4m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/35e219319ce840308767bbb67a3364cc : SUCCESS in 5m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/13da2086798f4fd6917c7798b6454b60 : FAILURE in 19m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5d7b4173172746839c28e8f191007154 : SUCCESS in 3m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bd708185f8514365b95e7352b26b5081 : FAILURE in 1h 07m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/49402c7256c24b808ea31278d100411e : FAILURE in 1h 39m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/30a3592bd27d431fb3c686447f644c7f : SUCCESS in 1h 59m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/399604e3b23b4ad598221d4a00ac3796 : SUCCESS in 1h 46m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3e9deebaf57c44b6b4ddceac06d0b740 : SUCCESS in 19m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/043c04ba617a41aba5e57baf90db4a5d : SUCCESS in 1h 13m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2267ea33f47d431d9b78c573c1db3fa1 : SUCCESS in 1h 35m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c4295b3bf3254c1593e73d574c8ef732 : SUCCESS in 27m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/07f043459d264195bb66c4e18e6a1a57 : SUCCESS in 2h 01m 25s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/974dcaba5bc0410d87af0baa64997864 : SUCCESS in 2h 02m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/12b20466798f431fb9378b7f442e9790 : SUCCESS in 1h 23m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea57e20ea7474e69b877026881bad86a : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":17},{"id":"83137b8c438ba36c9622babc443be592e6e35923","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-12-19 16:58:39.000000000","message":"Patch Set 17: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"2c36159c50b0c2500f47465ef8a2ef42b7154ed8","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-12-19 17:53:45.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"29102abe751ec934ace35d17ba78754929993acf","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-19 17:56:02.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/34ebea43b74e4f368beb2c026020fd42 : SUCCESS in 46m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d99d982123144b798d32aebc46f82adc : SUCCESS in 51m 38s","accounts_in_message":[],"_revision_number":17},{"id":"ebba3dc7f6801ade3e936e2cb48df38f3bb7fe8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 18:39:24.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a24e9984724475399fa73b0cfd937de\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c493185fc53547079dd42fe898e27f4a : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6c3a4929c7814ed986e98fcccd0c80d0 : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ad08462a86304004b40f04d1f132abe3 : SUCCESS in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"f8242a1c3d2127744dcd27e22bf05cb8a25ff663","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 20:12:04.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/665f098c02374566b8c18656ccdf4aed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13da2beea2374a98a531ef11db0ee055 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4ca28208f19483dabe214041b45bb29 : SUCCESS in 12m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3f2759ddaa94b508d9df04fb79242a9 : SUCCESS in 12m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/688bf255676c4171b829924e46fa7733 : SUCCESS in 11m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/88d6094c42304ec4873cfac7625e3e82 : SUCCESS in 1h 02m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/22dc6615d16b4b1783f0d8a52eec3acf : SUCCESS in 32m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b2106cc3021464887d95a238e4a1c30 : SUCCESS in 1h 44m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb97ef6143e340b5a951779c06ac2eaa : SUCCESS in 32m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e54139ff5c554418bc069ada1ca5ced5 : SUCCESS in 16m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e9ac87f8e4ce48829415714c8a9e1aa2 : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0e3fe5cb5c4f472599b4f781d9535725 : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6f00e5c8a6604b6cbc9d722b9526363c : SUCCESS in 3m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5a4c5476c288410aa3bbc2a583e46315 : SUCCESS in 8m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5afb874bcd03406d85e1f76ff9473175 : FAILURE in 40m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5b9512a0d25140ff8c0402f988e58fa6 : SUCCESS in 3m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/13226df033514c92bd22da05db2dd7ec : FAILURE in 18m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0ab05a9b671c4aa184623fc146e3bd70 : FAILURE in 2h 08m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/73eef5a48528495a99e00c4cc13fa5ea : SUCCESS in 1h 41m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/41e34c1f352c48ce8a0d2c15375f2bb5 : SUCCESS in 1h 31m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/695f51596aab4aaebb3045c8b2cee22b : SUCCESS in 36m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ea0f68ad6ebb410a85a252a9821d4642 : SUCCESS in 1h 18m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3fb62e60b6d242219b909c11186c9366 : SUCCESS in 1h 44m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0a9ec52a28b84669817d253a833ed2bb : SUCCESS in 33m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4527de48118b477b94dab3d2005d1dc0 : SUCCESS in 1h 26m 59s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/846ddcee98144acbb986ede6b2dc35c9 : SUCCESS in 1h 35m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/743b64c3b5ed4debb7e7ee705e329ac8 : SUCCESS in 1h 12m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc74c9af839143bb8d89879a10ff33e7 : SUCCESS in 41m 29s","accounts_in_message":[],"_revision_number":17},{"id":"22fd2df419302584f60ab58e014f4b8be75c2e03","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-12-19 20:18:29.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"13feb2aebde4378b11c655efafe6f93fd5b48f39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 20:37:48.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94609d85ff50410886e20142b1cd0dd9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bcec0c68940045ed9aeea2745e4d3764 : SUCCESS in 16m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/244d0de44314468887eb2b7a48a18d0c : SUCCESS in 17m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bf2b0ed1a6654958b03aca94a05a75d0 : SUCCESS in 13m 57s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"517eba3cf24cc04951c60b5458355dabb9af6f5b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-12-19 20:55:52.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-841315-17-fc : SUCCESS in 1h 26s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-841315-17-iscsi : SUCCESS in 1h 11m 20s","accounts_in_message":[],"_revision_number":17},{"id":"4a38e5e9934f727537009b47d9a8dcec5ce30927","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-19 22:13:34.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2611dd299538454c8e5d192acd6ba7df : SUCCESS in 53m 44s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/22458c77422141f48955f6314829dfa1 : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":17},{"id":"7b71e433929e307caaf4c41f8676a666a726536f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 22:37:55.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03db484ccfeb4f1e9f9fa5efd407ae6d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6634a98064cd4c9c97071b1a9485f4ca : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8507d1a394a34792951acffc9486587b : SUCCESS in 7m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ece11f3d1564faa8c758f0b27e32ca1 : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a943f568137a4eeb96331e2c917920fe : SUCCESS in 14m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/d58a996a760640889078ec41671a9e4b : SUCCESS in 41m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eef240a19b174ebba3b23f854a094621 : SUCCESS in 1h 00m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/168d250953a2445e9c66a15c0391cf16 : SUCCESS in 1h 36m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8bb419c2797741fb8696af810826ed81 : SUCCESS in 48m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/43106a0b4bbb4128bc0f05361d49d1d8 : SUCCESS in 25m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/305c030348b74bdcac6609df1772a504 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bf966ea0ea284a3c9b45458eb425922f : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/596ef273cc09450a957d2bb9b96f8f14 : SUCCESS in 3m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fd5b33d734164a79a933ceb6dc0f5c41 : SUCCESS in 9m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5b8927db47a24c81b0c2848c4e260302 : FAILURE in 37m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/369fca68cc3a4bfba7daf507a06a5e57 : SUCCESS in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b02fc59663384d44866c2a9b30d9ba96 : SUCCESS in 1h 08m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/802ad7a0e2a8449a90f5b730a495824c : FAILURE in 2h 14m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d8b6a9f4946e4d03be613f903772ffae : SUCCESS in 1h 53m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/713b66f994c54900be023aaead4a4b9b : SUCCESS in 1h 18m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1440ef575a534ddd8f9f867dd43fa769 : SUCCESS in 37m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/399bfafbe4ce4c2a89712380a27ca675 : SUCCESS in 1h 14m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a651a7bdc1254fe3a63b1f532e05a12e : SUCCESS in 1h 34m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/da22de26658d4705b98e4749d4608207 : SUCCESS in 40m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cf62b34360e74e958d05415bd46c5ce9 : SUCCESS in 2h 05m 48s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/3ecf39e6f04c4fefb957d3a26e8d3405 : SUCCESS in 1h 06m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d8a3ae13dc244fff93cc6e47fdb55af6 : SUCCESS in 1h 32m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0936b02a4e1b47d29d9e098b4a5c80c1 : SUCCESS in 55m 03s","accounts_in_message":[],"_revision_number":17},{"id":"6d496ee539eb6d38cff2fb1794172a6cca53689c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-12-19 22:44:38.000000000","message":"Patch Set 17:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/19_12_2025_22_38_02\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":17},{"id":"474eba032bf196986793b57c2902437dae72f2e9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-01-02 16:37:55.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"91e94704fecd8cb6ab27f39f51284d44cfaa3d23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 16:38:32.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"4f040b346918e4222ffd986da900e34e978664f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 18:48:22.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/137651dd2db242e6bab13a79ab445d19\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e9939d006744396afad7e9d646c5a04 : SUCCESS in 6m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e33d7ca21a74b42a38e37dfa09f3865 : SUCCESS in 10m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8f0165823c764e60a5c3d68f7b3fe7d1 : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/251b2ee88f1c4a8e99c39929f53d5d43 : SUCCESS in 10m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/1e771c337c1741348aa12157d82fdca8 : SUCCESS in 56m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fd64dac05ccf4405b162d29516a806d8 : SUCCESS in 45m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/65a412102e544de2853d0d3a0cd94744 : SUCCESS in 1h 13m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/422dc2a1fae94154a32aa6897706bfd0 : SUCCESS in 1h 03m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dcd6fe3a94f64a1ca5a3c8e864c437ac : SUCCESS in 1h 11m 44s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7ddafb169f914be48fae4012956daec7 : SUCCESS in 2h 02m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bda4476ac7b04e0583ccadb31be033c5 : SUCCESS in 45m 51s","accounts_in_message":[],"_revision_number":17},{"id":"7b0a4647e503e50d3193c15bbe860299fef70434","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 18:48:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"c53983d23cd2e99e5afaf25d32ccd891dc6b2191","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 18:50:05.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4ef5bbff77c40cebb855b686fbf2368\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbe4768c4abe4e6ebf9132985b2b53a7 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":17},{"id":"8f0ff1da65e081134a12abedd91008b288ad3705","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-01-02 19:35:59.000000000","message":"Patch Set 17:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/46/ : ABORTED","accounts_in_message":[],"_revision_number":17},{"id":"cb34e27996f34029d28f691036bcb25225b251ef","author":{"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},"date":"2026-01-03 00:43:57.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"28c523bf9aaf4bd402b5780ee468644b9b344340","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-03 01:55:56.000000000","message":"Patch Set 17:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_50//index.html : SUCCESS in 71m","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"028557388b0eecb5227876d9b0eaf5ba3fd1b7f6","revisions":{"c9e78876b6b62a05f0578503013af4c9af30b57c":{"kind":"REWORK","_number":1,"created":"2022-05-10 18:55:59.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/15/841315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/841315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/841315/1"}}},"commit":{"parents":[{"commit":"7b05eabac8a726b834e1e6e992bb1eff9486ca45","subject":"mypy: annotate remotefs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b05eabac8a726b834e1e6e992bb1eff9486ca45"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:49:32.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:55:08.000000000","tz":-240},"subject":"WIP: mypy: NFS","message":"WIP: mypy: NFS\n\nChange-Id: Ide947c8e3d3f050665c3acb804b6617d2b9c045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9e78876b6b62a05f0578503013af4c9af30b57c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9e78876b6b62a05f0578503013af4c9af30b57c"}]},"branch":"refs/heads/master"},"67fccc229f1a7ae505efbb9327fd91cbfd99b9f4":{"kind":"REWORK","_number":2,"created":"2022-07-27 14:36:18.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/15/841315/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/841315/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/841315/2"}}},"commit":{"parents":[{"commit":"d5058ed59879c88946abe3b727c7241dc5a0b5ec","subject":"mypy: work around mypy bug #13214","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5058ed59879c88946abe3b727c7241dc5a0b5ec"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:49:32.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-07-27 14:35:08.000000000","tz":-240},"subject":"WIP: mypy: NFS","message":"WIP: mypy: NFS\n\nChange-Id: Ide947c8e3d3f050665c3acb804b6617d2b9c045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/67fccc229f1a7ae505efbb9327fd91cbfd99b9f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/67fccc229f1a7ae505efbb9327fd91cbfd99b9f4"}]},"branch":"refs/heads/master"},"0c2ac4f152fdfa8916d40ae46dac83c282e562a8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-03-21 13:37:24.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/15/841315/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/841315/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/841315/3"}}},"commit":{"parents":[{"commit":"7489360d0c5b9bc12785eea951bae449ad92629d","subject":"Merge \"Lightbits update Qos doc\u0027s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7489360d0c5b9bc12785eea951bae449ad92629d"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:49:32.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-03-21 13:37:23.000000000","tz":-240},"subject":"WIP: mypy: NFS","message":"WIP: mypy: NFS\n\nChange-Id: Ide947c8e3d3f050665c3acb804b6617d2b9c045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0c2ac4f152fdfa8916d40ae46dac83c282e562a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0c2ac4f152fdfa8916d40ae46dac83c282e562a8"}]},"branch":"refs/heads/master"},"b64435b5cd62499d46e588a600b23d9573156e52":{"kind":"REWORK","_number":4,"created":"2025-03-21 14:13:13.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/15/841315/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/841315/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/841315/4"}}},"commit":{"parents":[{"commit":"7489360d0c5b9bc12785eea951bae449ad92629d","subject":"Merge \"Lightbits update Qos doc\u0027s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7489360d0c5b9bc12785eea951bae449ad92629d"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:49:32.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-03-21 14:09:54.000000000","tz":-240},"subject":"WIP: mypy: NFS","message":"WIP: mypy: NFS\n\nChange-Id: Ide947c8e3d3f050665c3acb804b6617d2b9c045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b64435b5cd62499d46e588a600b23d9573156e52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b64435b5cd62499d46e588a600b23d9573156e52"}]},"branch":"refs/heads/master"},"8dfdeb7f47a6f8414c9bfce752be20f914f63575":{"kind":"REWORK","_number":5,"created":"2025-03-25 13:50:42.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/15/841315/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/841315/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/841315/5"}}},"commit":{"parents":[{"commit":"1cd506f56b7755b22ef770c22539bf8202854d1b","subject":"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1cd506f56b7755b22ef770c22539bf8202854d1b"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:49:32.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-03-25 13:50:13.000000000","tz":-240},"subject":"WIP: mypy: NFS","message":"WIP: mypy: NFS\n\nChange-Id: Ide947c8e3d3f050665c3acb804b6617d2b9c045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8dfdeb7f47a6f8414c9bfce752be20f914f63575"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8dfdeb7f47a6f8414c9bfce752be20f914f63575"}]},"branch":"refs/heads/master"},"6f9c6a932e6effb05e0c57e9affc288e224b391c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-03-27 14:54:49.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/15/841315/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/841315/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/841315/6"}}},"commit":{"parents":[{"commit":"625946e14869cd9812cacb4fe5a4112506185fea","subject":"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/625946e14869cd9812cacb4fe5a4112506185fea"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:49:32.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-03-27 14:54:49.000000000","tz":0},"subject":"WIP: mypy: NFS","message":"WIP: mypy: NFS\n\nChange-Id: Ide947c8e3d3f050665c3acb804b6617d2b9c045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6f9c6a932e6effb05e0c57e9affc288e224b391c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6f9c6a932e6effb05e0c57e9affc288e224b391c"}]},"branch":"refs/heads/master","description":"Rebase"},"fbf7de7276b385bbf742ce65b914f2d2918d503a":{"kind":"REWORK","_number":7,"created":"2025-03-27 15:09:58.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/15/841315/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/841315/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/841315/7"}}},"commit":{"parents":[{"commit":"625946e14869cd9812cacb4fe5a4112506185fea","subject":"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/625946e14869cd9812cacb4fe5a4112506185fea"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:49:32.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-03-27 15:09:40.000000000","tz":-240},"subject":"WIP: mypy: NFS","message":"WIP: mypy: NFS\n\nChange-Id: Ide947c8e3d3f050665c3acb804b6617d2b9c045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fbf7de7276b385bbf742ce65b914f2d2918d503a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fbf7de7276b385bbf742ce65b914f2d2918d503a"}]},"branch":"refs/heads/master"},"f0a13010e068ed2f231f5e0e65db72a1b81faf38":{"kind":"REWORK","_number":8,"created":"2025-03-31 14:40:05.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/15/841315/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/841315/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/841315/8"}}},"commit":{"parents":[{"commit":"b3e955bd2c16f719f93ef91a6875cede866cb3c4","subject":"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b3e955bd2c16f719f93ef91a6875cede866cb3c4"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:49:32.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-03-31 14:39:20.000000000","tz":-240},"subject":"mypy: NFS driver","message":"mypy: NFS driver\n\nChange-Id: Ide947c8e3d3f050665c3acb804b6617d2b9c045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f0a13010e068ed2f231f5e0e65db72a1b81faf38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f0a13010e068ed2f231f5e0e65db72a1b81faf38"}]},"branch":"refs/heads/master"},"851f3de1f9cfa9dc0bd3f43318afa0ca41ce7083":{"kind":"REWORK","_number":9,"created":"2025-04-24 14:50:13.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/15/841315/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/841315/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/841315/9"}}},"commit":{"parents":[{"commit":"541531746c46d1e58237606f9061048ceaeee20f","subject":"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/541531746c46d1e58237606f9061048ceaeee20f"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:49:32.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-04-24 14:48:29.000000000","tz":-240},"subject":"mypy: NFS driver","message":"mypy: NFS driver\n\nChange-Id: Ide947c8e3d3f050665c3acb804b6617d2b9c045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/851f3de1f9cfa9dc0bd3f43318afa0ca41ce7083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/851f3de1f9cfa9dc0bd3f43318afa0ca41ce7083"}]},"branch":"refs/heads/master"},"a1d8cb2c40f917c11734fb1677b22803405d607a":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-06-26 11:38:03.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/15/841315/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/841315/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/841315/10"}}},"commit":{"parents":[{"commit":"d75cf8769b078dc3ea40eef528b8e2ef77afc5d0","subject":"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d75cf8769b078dc3ea40eef528b8e2ef77afc5d0"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:49:32.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-06-26 11:38:03.000000000","tz":0},"subject":"mypy: NFS driver","message":"mypy: NFS driver\n\nChange-Id: Ide947c8e3d3f050665c3acb804b6617d2b9c045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1d8cb2c40f917c11734fb1677b22803405d607a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1d8cb2c40f917c11734fb1677b22803405d607a"}]},"branch":"refs/heads/master","description":"Rebase"},"edd1d77d440523796820f9c0d6aad82fd6c8299f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":11,"created":"2025-07-01 16:41:02.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/15/841315/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/841315/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/841315/11"}}},"commit":{"parents":[{"commit":"08a71933934c430c3163f18a8ae8b9657e309694","subject":"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08a71933934c430c3163f18a8ae8b9657e309694"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:49:32.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-01 16:39:02.000000000","tz":-240},"subject":"mypy: NFS driver","message":"mypy: NFS driver\n\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\nChange-Id: Ide947c8e3d3f050665c3acb804b6617d2b9c045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/edd1d77d440523796820f9c0d6aad82fd6c8299f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/edd1d77d440523796820f9c0d6aad82fd6c8299f"}]},"branch":"refs/heads/master"},"882c16240282ceeb8e8189cf6745a67e7a3d3a4e":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-07-23 12:43:39.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/15/841315/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/841315/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/841315/12"}}},"commit":{"parents":[{"commit":"d9d5ce5c5a0422e6d590fd5df7f3992c53bd9a99","subject":"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d9d5ce5c5a0422e6d590fd5df7f3992c53bd9a99"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:49:32.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-23 12:43:39.000000000","tz":0},"subject":"mypy: NFS driver","message":"mypy: NFS driver\n\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\nChange-Id: Ide947c8e3d3f050665c3acb804b6617d2b9c045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/882c16240282ceeb8e8189cf6745a67e7a3d3a4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/882c16240282ceeb8e8189cf6745a67e7a3d3a4e"}]},"branch":"refs/heads/master","description":"Rebase"},"7bce64eeededb70b67fe30c1c624b8da58e6be17":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-09-29 18:35:09.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/15/841315/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/841315/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/841315/13"}}},"commit":{"parents":[{"commit":"ac6f4df0c9c88b0386fb9e2d1827ebc3d9b8712a","subject":"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac6f4df0c9c88b0386fb9e2d1827ebc3d9b8712a"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:49:32.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-09-29 18:35:09.000000000","tz":0},"subject":"mypy: NFS driver","message":"mypy: NFS driver\n\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\nChange-Id: Ide947c8e3d3f050665c3acb804b6617d2b9c045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7bce64eeededb70b67fe30c1c624b8da58e6be17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7bce64eeededb70b67fe30c1c624b8da58e6be17"}]},"branch":"refs/heads/master","description":"Rebase"},"0818b1024362ff699f9621db57f022606dc2e56c":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-10-09 15:10:22.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/15/841315/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/841315/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/841315/14"}}},"commit":{"parents":[{"commit":"28e8f07bc52f0d1cf577e2d9a12c957dfbb9d2f7","subject":"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/28e8f07bc52f0d1cf577e2d9a12c957dfbb9d2f7"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:49:32.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-10-09 15:10:22.000000000","tz":0},"subject":"mypy: NFS driver","message":"mypy: NFS driver\n\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\nChange-Id: Ide947c8e3d3f050665c3acb804b6617d2b9c045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0818b1024362ff699f9621db57f022606dc2e56c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0818b1024362ff699f9621db57f022606dc2e56c"}]},"branch":"refs/heads/master","description":"Rebase"},"c0ff7e331a4f521b38bd6c4a421a1f7c116adb9f":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2025-11-25 14:43:50.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/15/841315/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/841315/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/841315/15"}}},"commit":{"parents":[{"commit":"4a84a60be3d1c144e36d5c9ed3d4d82aff8a7fbe","subject":"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4a84a60be3d1c144e36d5c9ed3d4d82aff8a7fbe"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:49:32.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-11-25 14:43:50.000000000","tz":0},"subject":"mypy: NFS driver","message":"mypy: NFS driver\n\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\nChange-Id: Ide947c8e3d3f050665c3acb804b6617d2b9c045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0ff7e331a4f521b38bd6c4a421a1f7c116adb9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0ff7e331a4f521b38bd6c4a421a1f7c116adb9f"}]},"branch":"refs/heads/master","description":"Rebase"},"59b32d49e8275993007dd421aba44ff3adeeceb6":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2025-12-19 14:42:44.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/15/841315/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/841315/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/841315/16"}}},"commit":{"parents":[{"commit":"c68475a3d0dd54812f3618962c940fd70fd4b381","subject":"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c68475a3d0dd54812f3618962c940fd70fd4b381"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:49:32.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-12-19 14:42:44.000000000","tz":0},"subject":"mypy: NFS driver","message":"mypy: NFS driver\n\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\nChange-Id: Ide947c8e3d3f050665c3acb804b6617d2b9c045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/59b32d49e8275993007dd421aba44ff3adeeceb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/59b32d49e8275993007dd421aba44ff3adeeceb6"}]},"branch":"refs/heads/master","description":"Rebase"},"028557388b0eecb5227876d9b0eaf5ba3fd1b7f6":{"kind":"REWORK","_number":17,"created":"2025-12-19 14:47:54.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/15/841315/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/841315/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/841315/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/841315/17"}}},"commit":{"parents":[{"commit":"c68475a3d0dd54812f3618962c940fd70fd4b381","subject":"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c68475a3d0dd54812f3618962c940fd70fd4b381"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-05-10 18:49:32.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-12-19 14:47:38.000000000","tz":-300},"subject":"mypy: NFS driver","message":"mypy: NFS driver\n\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\nChange-Id: Ide947c8e3d3f050665c3acb804b6617d2b9c045c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/028557388b0eecb5227876d9b0eaf5ba3fd1b7f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/028557388b0eecb5227876d9b0eaf5ba3fd1b7f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
