)]}'
{"id":"openstack%2Fcinder~843018","triplet_id":"openstack%2Fcinder~master~Ia685e398e5c41ae36e1a37530f0a742f79533e9f","project":"openstack/cinder","branch":"master","topic":"bug/1969531","attention_set":{"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-03-23 09:31:25.000000000","reason":"\u003cGERRIT_ACCOUNT_38908\u003e replied on the change","reason_account":{"_account_id":38908,"name":"Deepa Reddy","email":"Deepa.Reddy@netapp.com"}},"34489":{"account":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"last_update":"2022-05-26 14:44:07.000000000","reason":"Someone else replied on the change"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-03-23 09:31:15.000000000","reason":"\u003cGERRIT_ACCOUNT_38908\u003e replied on the change","reason_account":{"_account_id":38908,"name":"Deepa Reddy","email":"Deepa.Reddy@netapp.com"}},"38623":{"account":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"last_update":"2026-03-23 09:31:15.000000000","reason":"\u003cGERRIT_ACCOUNT_38908\u003e replied on the change","reason_account":{"_account_id":38908,"name":"Deepa Reddy","email":"Deepa.Reddy@netapp.com"}}},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2022-06-02 06:31:12.000000000","reason":"Raghavendra Tilay replied on the change"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-03-25 06:11:00.000000000","reason":"\u003cGERRIT_ACCOUNT_38059\u003e replied on the change","reason_account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}}},"hashtags":[],"change_id":"Ia685e398e5c41ae36e1a37530f0a742f79533e9f","subject":"Fix NetApp NFS storage migration between backends","status":"NEW","created":"2022-05-23 19:25:28.000000000","updated":"2026-04-08 13:46:21.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":12,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"fb2e6220ae45682afc9be0dd7d1205d9ecb282ac","_number":843018,"virtual_id_number":843018,"owner":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-27 15:39:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"disliked":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":-1,"date":"2023-07-04 13:33:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},{"value":1,"date":"2022-06-02 06:31:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":1,"date":"2022-07-06 16:48:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":-1,"date":"2026-03-25 06:11:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38908,"name":"Deepa Reddy","email":"Deepa.Reddy@netapp.com"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-23 19:25:37.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-23 19:47:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-23 19:50:04.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-05-23 20:26:54.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-05-23 20:37:45.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-05-23 20:55:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-23 21:03:38.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-23 21:49:37.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-23 22:59:46.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-24 00:42:28.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-05-24 04:54:39.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-05-24 14:36:05.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-05-24 14:46:55.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-24 15:48:19.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-24 15:54:49.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-05-24 16:07:44.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-05-24 16:09:56.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-05-24 16:45:11.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-05-24 16:48:52.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-24 17:11:02.000000000","updated_by":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"reviewer":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"state":"REVIEWER"},{"updated":"2022-05-24 20:35:33.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-05-24 22:44:08.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-05-24 23:16:19.000000000","updated_by":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"reviewer":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"state":"CC"},{"updated":"2022-05-24 23:16:51.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-05-25 00:58:18.000000000","updated_by":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"reviewer":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"state":"REVIEWER"},{"updated":"2022-05-25 01:55:44.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-05-25 07:16:23.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2022-05-25 19:24:26.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-05-27 15:57:37.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-06-02 06:31:12.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2022-06-23 09:41:19.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-06 16:48:18.000000000","updated_by":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"reviewer":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"state":"REVIEWER"},{"updated":"2023-07-04 13:33:59.000000000","updated_by":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"reviewer":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"state":"REVIEWER"},{"updated":"2026-03-23 09:31:15.000000000","updated_by":{"_account_id":38908,"name":"Deepa Reddy","email":"Deepa.Reddy@netapp.com"},"reviewer":{"_account_id":38908,"name":"Deepa Reddy","email":"Deepa.Reddy@netapp.com"},"state":"CC"},{"updated":"2026-03-23 09:31:15.000000000","updated_by":{"_account_id":38908,"name":"Deepa Reddy","email":"Deepa.Reddy@netapp.com"},"reviewer":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-03-23 09:31:15.000000000","updated_by":{"_account_id":38908,"name":"Deepa Reddy","email":"Deepa.Reddy@netapp.com"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-03-23 09:31:15.000000000","updated_by":{"_account_id":38908,"name":"Deepa Reddy","email":"Deepa.Reddy@netapp.com"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-03-23 09:31:25.000000000","updated_by":{"_account_id":38908,"name":"Deepa Reddy","email":"Deepa.Reddy@netapp.com"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"}],"messages":[{"id":"358f7b78b3a7fe053b3b7002c0fb79a540717092","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"date":"2022-05-23 19:25:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e27db598da4917366da3ee3f7176269b1493acc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-23 19:25:37.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16061/","accounts_in_message":[],"_revision_number":1},{"id":"f91b9798db96249d2541ae53e61c2e934c44a483","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-23 19:27:49.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16061/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"e3c3c73a96529cd10309e414b292c72f56f692ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 19:47:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eecc2e9bed1046adb2b69b0685fec653 : SUCCESS in 13m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bff629e3afa846c89f9e2bd078af8161 : SUCCESS in 18m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34824bd248ff1ce80a9d44ca0937d3d492136d4d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-23 19:50:04.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1653333936754 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"0e6f0ca01f04eddd5700ebe02aab025f846085f7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-23 20:26:54.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"7cd671483e6137d65bcd166d27509ae2f2e35fff","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-23 20:37:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/843018/1/iscsi/ : SUCCESS in 1h 09m 08s\n- cinder-smb http://cloudbase-ci.com/cinder/843018/1/smb/ : SUCCESS in 1h 11m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8a1c73b48aac96f0006591a44837b3f855eee23a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 20:55:06.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b66bb1f339d34d31b639c422484724c0 : FAILURE in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/686d6b177fbf45558fee229c07567f64 : FAILURE in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/690bcc1fa81e4d5e975cf1a670d47974 : SUCCESS in 11m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d6afc1fdf4564e1c8c4db71ab09579f8 : SUCCESS in 11m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dc97f80f8d64bbe9fcfc423cac625cd : SUCCESS in 11m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/533da38785074168bd98dc6804416781 : SUCCESS in 53m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/053d385b48f2482e9e54325b5db1e380 : SUCCESS in 1h 07m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a7feeebe9a348929e56e38df9e0c5de : SUCCESS in 48m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f1ca46da655d4ced841c0a191d576e80 : FAILURE in 3m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4feaef3e9203427f85c85a7de69b58db : SUCCESS in 6m 21s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f44c6fdf0576489c8ad7b395ed5753fb : FAILURE in 3m 37s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3261ae566f4248568adf7a5a9e253cef : SUCCESS in 7m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d66baa5812e045cab72db7df9aaeb471 : FAILURE in 3m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c57b016757fc49fdacfaaa576dc2e794 : SUCCESS in 38m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/abb61d7e5b9942848f4ee8cd31eb4736 : FAILURE in 3m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cadcd02ad87041e9abf503363c873666 : SUCCESS in 1h 10m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0da413a75b6c4ef1ad463e4a28541cf4 : POST_FAILURE in 13m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d6caf0a32cc94d30946e67dd85ba6563 : SUCCESS in 1h 27m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f55470116b9f47d2833838cc324dec75 : FAILURE in 14m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/13f6aee0fcb549b7bf4bdf96e0db5d0e : SUCCESS in 1h 23m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a7f3a11b4d544d5f94dda564fe4afb8c : FAILURE in 9m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4e41ad25916440ff8f6c88d7b05fe901 : SUCCESS in 1h 16m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/19520fbc1aca4480b91c1113b9beb530 : FAILURE in 11m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13ce0d9cfeea4cd8a68a0a6de9b5c1b3 : SUCCESS in 48m 37s","accounts_in_message":[],"_revision_number":1},{"id":"8490d923a03c06378ce28e01ba8aba2eec08042e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-23 21:03:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/02a9cb12330d4842a9410370e72ffbcb : RETRY_LIMIT in 11m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6a6d02ea44b86c94430c88b337f5a74254df2b17","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-23 21:49:37.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/843018/1/check/tempest-dsvm-full-focal-py3/e4f0a28/ : FAILURE in 2h 20m 31s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/843018/1/check/devstack-plugin-ceph-dsvm-focal-py3/088f2e4/ : SUCCESS in 2h 12m 17s","accounts_in_message":[],"_revision_number":1},{"id":"da71c1a25519a1274a68a6c86a64bae5c3de80ef","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-23 22:59:46.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-843018-1/ns5_nfs : FAILURE in 3h 32m 14s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-843018-1/ns5_iscsi : SUCCESS in 3h 33m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"7952eefe0488cccebd52e646109886df3ee9e4fd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-24 00:42:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/18/843018/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/40c4fe4/ : TIMED_OUT in 3h 02m 53s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/18/843018/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7e3cbed/ : FAILURE in 1h 10m 46s","accounts_in_message":[],"_revision_number":1},{"id":"66c54d954fa590c6797dbe9e53bd7780352f3b94","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-24 04:54:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/18/843018/1/check/iscsi-driver-nimble/1302baf/1302baf.html : SUCCESS in 1h 37m 54s","accounts_in_message":[],"_revision_number":1},{"id":"667bb1d2c2c2c8bde7663c17008f6a0cd2f63bec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"date":"2022-05-24 12:14:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"429a210c425d4ec4632875ea4962629d04713e5d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-24 12:14:44.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16064/","accounts_in_message":[],"_revision_number":2},{"id":"09743e951e6bba0524c5c1ad2facd7102393c551","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-24 12:17:04.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16064/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"731c952fde9805e47fcf09acf3b0ff03f4c983c2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-24 12:40:42.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1653394483390 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"0158ec0554419f864310411a46710c72dddd12f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 12:42:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d52803bd53548729ddd1575400bcfd2 : SUCCESS in 21m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c9b53b9b9704237a0a7a9edc87ddd71 : SUCCESS in 23m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"434e9899cc5f0db7a49d2ebd5066700e9fcbbbfd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-24 12:53:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b41572b679cd4a24bf79800ab94d2b46 : RETRY_LIMIT in 11m 26s","accounts_in_message":[],"_revision_number":2},{"id":"1ca79544d3d23b5e97b29541195e6ef0b7b9d63a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-24 13:50:03.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/843018/2/iscsi/ : FAILURE in 1h 34m 53s\n- cinder-smb http://cloudbase-ci.com/cinder/843018/2/smb/ : SUCCESS in 1h 00m 33s","accounts_in_message":[],"_revision_number":2},{"id":"f3e7cf29e7bb4feba8f6884b6f728f27d826bfc6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-24 13:53:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/18/843018/2/check/iscsi-driver-nimble/5b85c4c/5b85c4c.html : SUCCESS in 1h 37m 13s","accounts_in_message":[],"_revision_number":2},{"id":"5367afa0254503748d821c9b06fedf954a52fd90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 14:35:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a612d74c5a7c4b929aa0f43ae093d299 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9630a4910e784aff8cce98601916378c : SUCCESS in 12m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3849a5472274e098b77e96ea4d5c568 : SUCCESS in 11m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbd3efcab74e4a71a76b4afd43896e00 : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efcd5b3fb8f3464f902627207b488102 : SUCCESS in 13m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/cdf791826fe0444aa6c6039f2e8cef20 : SUCCESS in 53m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eba153eef3834232b75a1fa066a5848f : SUCCESS in 1h 21m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51db2aed7f764168b9cbbd3a94df1ed4 : SUCCESS in 47m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/293f20dde3b24df3b435dc153eb7f82c : SUCCESS in 14m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fd7cfd8702344208bbcdbc7d8b83af89 : SUCCESS in 6m 56s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8b316ee5034f4360b3c86687a434a03a : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4a14bb8704a8491c945b04ec347599aa : SUCCESS in 7m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d208f046d3af43f281960068565f99a2 : SUCCESS in 6m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4d57c98b1035456b846622425850682e : SUCCESS in 39m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4c9536b4c28845e0987df202f6869ec9 : FAILURE in 7m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/45591cbd1500477fb3d3aa0fcd26f335 : SUCCESS in 1h 14m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bf06af4c46544158867e28c0f856c42a : FAILURE in 1h 34m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b5e8d8d7433c49cfa1eb1d1677b59cae : SUCCESS in 1h 30m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ae37e7fb1c4b41158c0cbeafc98cdb7f : SUCCESS in 55m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2eeeff2199364627b0fa5e22e6b4e162 : SUCCESS in 1h 17m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/92bcfa4068924c949be7142147764dec : SUCCESS in 40m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/be85fee654594060a3f37d922a2e9a03 : SUCCESS in 1h 22m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e5e66171e64646d295450ac09a068cc9 : SUCCESS in 2h 17m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f298f7d1c47f47c38efef40a749a8a34 : SUCCESS in 47m 23s","accounts_in_message":[],"_revision_number":2},{"id":"0ec5e97bb4e24694275d3b4ff64f72e5c8e1ddbd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-24 14:36:05.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"5c6f2904fc3f4deb2572cadc32f5cf9974283059","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-24 14:46:07.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/843018/2/check/tempest-dsvm-full-focal-py3/e93f7c9/ : SUCCESS in 1h 37m 16s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/843018/2/check/devstack-plugin-ceph-dsvm-focal-py3/67b2fe3/ : FAILURE in 2h 29m 09s","accounts_in_message":[],"_revision_number":2},{"id":"7f25503dd4b7d0b495d05b3ac8e6fd8401151f63","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-24 14:46:55.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/18/843018/2/check/DellEMC_VNX_iSCSI/14df26c/DellEMC_VNX_iSCSI/444 : FAILURE in 1s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/18/843018/2/check/DellEMC_VNX_FC/fc0f8ce/DellEMC_VNX_FC/463 : FAILURE in 10m 40s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"2a4206c2fdea9d892976bfd127d1623c2727370b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-24 15:19:35.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/18/843018/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/1ab97c8/ : TIMED_OUT in 3h 03m 10s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/18/843018/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d47c891/ : TIMED_OUT in 3h 03m 14s","accounts_in_message":[],"_revision_number":2},{"id":"b9a01f8266cfcbc2c4c0021974b26755016c471f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-24 15:47:10.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-843018-2/ns5_nfs : SUCCESS in 3h 29m 48s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-843018-2/ns5_iscsi : SUCCESS in 3h 32m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"f60d5431e014cb74e4d1281c6ba194ffdf2cac5a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-24 15:48:19.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/18/843018/2/check/DellEMC_PowerFlex/c2dab5d/DellEMC_PowerFlex/787 : SUCCESS in 1h 11m 57s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"d85ee65aa7d5f4d0452831e7f7d0b64c49f08218","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-05-24 15:54:49.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/18/843018/2/check/DellEMC_PowerStore_iSCSI/0858c89/DellEMC_PowerStore_iSCSI/532 : SUCCESS in 1h 18m 30s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/18/843018/2/check/DellEMC_PowerStore_FC/c803f08/DellEMC_PowerStore_FC/557 : SUCCESS in 1h 08m 37s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"d8712eda2184b100c9fd01f94bb8d67e4d4b4cdd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-05-24 16:07:44.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/843018/2/upstream-check/cinder-tempest-plugin-ontap-fcp/fdce8d7/ : SUCCESS in 1h 03m 33s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/843018/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/c41bc52/ : FAILURE in 1h 05m 59s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/843018/2/upstream-check/cinder-tempest-plugin-ontap-nfs/dd1c69e/ : FAILURE in 1h 30m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"f08baf5c8e038b49b3696cfb6c5de6ad6881d098","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-05-24 16:09:56.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/18/843018/2/check/DellEMC_SC_FC/6c6d7ad/DellEMC_SC_FC/634 : SUCCESS in 44m 47s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/18/843018/2/check/DellEMC_SC_iSCSI/3dd1628/DellEMC_SC_iSCSI/655 : SUCCESS in 1h 33m 44s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"32ed3f19f61632e4ca6bd4706495cf8c7c59eec1","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-05-24 16:45:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/843018-2/HPMSA-iSCSI/ : SUCCESS in 52:14.00\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/843018-2/HPMSA-FC/ : SUCCESS in 45:30.12","accounts_in_message":[],"_revision_number":2},{"id":"62846d2dad46470b673e12a35ffc5fdeb59cca7a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-05-24 16:48:52.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"31e75a07474b3ab0930f88e3944732d44370f788","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2022-05-24 17:11:02.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"421862779e57070e119668fa32c6fe63bf8ae37f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-24 20:35:33.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/843018/2/20220525031714/NEC-V-FC : SUCCESS in 1h 55m 32s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/843018/2/20220525031714/NEC-V-ISCSI : SUCCESS in 2h 18m 15s","accounts_in_message":[],"_revision_number":2},{"id":"dec097886d4cba034a6e903886be2c430fd1bea8","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-05-24 22:44:08.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/843018-2/PVME-iSCSI/ : SUCCESS in 1:07:18\n- PVME-FC https://os-logs.seagate.com/dell-ci/843018-2/PVME-FC/ : SUCCESS in 1:03:55","accounts_in_message":[],"_revision_number":2},{"id":"fa96ded587952cfcb9ac81bfb9b8c4381e03a5a1","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2022-05-24 23:16:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4c373d0cb13738766ba116e5bcb5b31b2f5a2d3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-24 23:16:51.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/843018/2/check/DellEMC_Unity_FC/348d7d5/DellEMC_Unity_FC/830 : SUCCESS in 1h 06m 34s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/843018/2/check/DellEMC_Unity_iSCSI/faf9a0f/DellEMC_Unity_iSCSI/796 : ABORTED in 3h 00m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"0748495ddb09fb48bc5068cbf5bc72f394eac078","author":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"date":"2022-05-25 00:58:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ae0fbeda228ae79b4ecf7bef4c35034bda8895c5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-25 00:59:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e259d3bbdac045dcaee320677e8d8aa9 : SUCCESS in 1h 25m 46s","accounts_in_message":[],"_revision_number":2},{"id":"7ffb1e52eaf1d58582d46b2f9b8953b0c4910359","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-05-25 01:55:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/843018-2/Seagate-iSCSI/ : FAILURE in 1:33:21\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/843018-2/Seagate-FC/ : SUCCESS in 1:11:38","accounts_in_message":[],"_revision_number":2},{"id":"bb55320d140b0a32a9119269e6a3ddb8424fc36e","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-05-25 07:16:23.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"385db56a5c08382c01a0241f500a792a52ded085","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-25 19:24:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/843018-2/Lenovo-iSCSI/ : SUCCESS in 1:07:37\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/843018-2/Lenovo-FC/ : SUCCESS in 1:05:51","accounts_in_message":[],"_revision_number":2},{"id":"5453b22209f02b5146280389a6b8e66ed8edb8e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"date":"2022-05-26 14:14:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d8fc934b21b7218c3d0d313f6414b7244801d933","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-26 14:14:27.000000000","message":"Patch Set 3:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16080/","accounts_in_message":[],"_revision_number":3},{"id":"03514a291d513c84f3b299e5e29115584b8cedbe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-26 14:16:32.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16080/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"fcc433673a2c656d182f36927c034045c063852b","author":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"date":"2022-05-26 14:19:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b95f1faf9aa1ef72c4484ecd66ef92392ffd44b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 14:27:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0fd5e7c7e8454ee9bdd4e7a9ca1fa5ed : FAILURE in 7m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d3b05c3730b74faebf91e25a14f48e1a : FAILURE in 7m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5df87567228c817e48a03cb154894e1c0fecedf1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-26 14:44:07.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1653574464498 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"53bd3110cc78cda8781841eaabeb5d334e7018d2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-26 15:33:37.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/843018/3/iscsi/ : SUCCESS in 1h 12m 55s\n- cinder-smb http://cloudbase-ci.com/cinder/843018/3/smb/ : SUCCESS in 1h 18m 40s","accounts_in_message":[],"_revision_number":3},{"id":"665de0ede47d55c1e0b19ba02096be44bd9c6928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 16:40:22.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18fc116ab778489eadcdb09f2aa3a911 : FAILURE in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dfe81c7f51e4cf59ac1f4a494b8e191 : FAILURE in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da327279dffa4d9b8bcb8d7e6dcfcbc9 : FAILURE in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d544faf705504fe284b8b81b3152ed3d : FAILURE in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eae687b48604c19af9eb51fbffd1d84 : SUCCESS in 11m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/40fc74e19b484075a9533fd9b5c469fa : SUCCESS in 1h 08m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1a5071b4861f4852bdcacda0883f80f2 : SUCCESS in 1h 18m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8170467641f4109854a9820e60f1667 : SUCCESS in 54m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b48f919e74f04d59b0c44cf012a2191f : FAILURE in 6m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/27f295ede7c34b9dad114f67cfc335ef : SUCCESS in 6m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ab120efd97424e7e8e8f7fba34c3026e : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fa7005ab4e6d4b26b3df9a49cd1df753 : SUCCESS in 7m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1accb1514b6c4c6c87c35bce832226d2 : SUCCESS in 6m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/393ec2ea819640b5b5981047d0ec08f5 : TIMED_OUT in 2h 14m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/02724cf3ca714bd2916103a985008e6b : FAILURE in 6m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/51de61a23fb14027a05bd61fb75fdcc9 : SUCCESS in 1h 30m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ec6d23acce284379985ae1eab3e9ca99 : FAILURE in 1h 21m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/aa124513feca4fce80fd4aeba85dde0d : SUCCESS in 2h 13m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/025af32273374845ab722a72a4e46d06 : FAILURE in 35m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d3f0636a75d9462490a8e860757a6511 : SUCCESS in 1h 23m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f50fae8e47d24185a4fe42c3ee0ec0d1 : SUCCESS in 39m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/01033033a82042e79635a7dcebd59bfe : SUCCESS in 1h 27m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d3f6519e480a4524876d5d3e9e2b556e : SUCCESS in 2h 21m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/102c57c43aa8462790da889127f70405 : SUCCESS in 42m 26s","accounts_in_message":[],"_revision_number":3},{"id":"e9990e7707f702d162b1087c25d3e0f1cc1abe98","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-26 16:40:58.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/18/843018/3/check/dsvm-tempest-cisco-zonemanager-job/c32880d/ : SUCCESS in 1h 07m 56s","accounts_in_message":[],"_revision_number":3},{"id":"7fa6ef5ba7262ac2879c2e200478ae2dd9e2f5a4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-26 17:20:43.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/18/843018/3/check/iscsi-driver-nimble/b03d40a/b03d40a.html : FAILURE in 1h 01m 10s","accounts_in_message":[],"_revision_number":3},{"id":"75a38242280ba3af51609b66fe95aa25d22d62ca","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-26 18:10:24.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/843018/3/check/tempest-dsvm-full-focal-py3/8aee396/ : SUCCESS in 2h 03m 38s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/843018/3/check/devstack-plugin-ceph-dsvm-focal-py3/744d329/ : TIMED_OUT in 3h 17m 19s","accounts_in_message":[],"_revision_number":3},{"id":"a1ba8e55e1db9a041ac5b9a279af4577a624ac70","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-26 19:17:48.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-843018-3/ns5_nfs : SUCCESS in 3h 38m 37s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-843018-3/ns5_iscsi : SUCCESS in 3h 46m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"cbaaf800cdc94fa392f1f55bd2cfa7924717de28","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-26 19:35:29.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/18/843018/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b2e616a/ : TIMED_OUT in 3h 03m 33s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/18/843018/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/1684194/ : FAILURE in 1h 23m 51s","accounts_in_message":[],"_revision_number":3},{"id":"2a4a425119244daeec2a64406daec5484a239f0a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-27 07:13:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/843018/3/20220527135540/NEC-V-FC : SUCCESS in 1h 53m 31s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/843018/3/20220527135540/NEC-V-ISCSI : SUCCESS in 2h 17m 28s","accounts_in_message":[],"_revision_number":3},{"id":"2d36ecc4f67215a89468fe95724e95410f94dcaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"date":"2022-05-27 12:56:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1fa2675ce6bc8a01488962d99a29e5e8c0bc680e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-27 12:56:32.000000000","message":"Patch Set 4:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16089/","accounts_in_message":[],"_revision_number":4},{"id":"68cb818ab07db01f32c26bf6db53e45c059b8774","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-27 12:56:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/843018/4/check/DellEMC_Unity_FC/f21e38f/DellEMC_Unity_FC/870 : FAILURE in 1s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/843018/4/check/DellEMC_Unity_iSCSI/6ade8cf/DellEMC_Unity_iSCSI/815 : FAILURE in 6s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"bcb5c6cf08def42fcb31213961d05bb9a10cf02b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-27 12:58:44.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16089/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"ed153990708ff162c78c83faae65b840e814ff6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 13:17:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/17008eb218904a4ca439f03a1f1f728b : SUCCESS in 18m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ec8a419cce9451a87d3ce54f76d81d1 : SUCCESS in 18m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"000edcaa631fb4c908d50c3b7a6ed5b2e14d2d76","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-27 13:22:44.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1653656192601 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"899bd98bed05a20a9bd55aab7830a0bc475821ac","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-27 13:57:37.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/18/843018/4/check/iscsi-driver-nimble/ffeb8d7/ffeb8d7.html : FAILURE in 1h 00m 09s","accounts_in_message":[],"_revision_number":4},{"id":"832441f3e562fdfc82bceaa065b95314e01e49cf","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-27 14:23:37.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/843018/4/iscsi/ : SUCCESS in 1h 26m 42s\n- cinder-smb http://cloudbase-ci.com/cinder/843018/4/smb/ : SUCCESS in 1h 24m 18s","accounts_in_message":[],"_revision_number":4},{"id":"fe3a0f43f1f90ce0521a217bb4732d83acbf9506","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-27 15:28:26.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/843018/4/check/tempest-dsvm-full-focal-py3/014ddb1/ : SUCCESS in 1h 44m 27s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/843018/4/check/devstack-plugin-ceph-dsvm-focal-py3/ab29fdd/ : SUCCESS in 2h 26m 07s","accounts_in_message":[],"_revision_number":4},{"id":"0df8520e03b50d74e1d050ee7f99e8859bf45c40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 15:39:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43affcb20ede4aa1abeb68ae8ba99f81 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c1a7e7adf72435db289b559e5bfd2d8 : SUCCESS in 9m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91bb47132f7a48728f4ddb5080895749 : SUCCESS in 10m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3b0887729284bbb9f2ce001cf5cf0ce : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cab3d427f0a74c68bc5c7ee2422d2708 : SUCCESS in 12m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/042328bd296042f48e9281ff2594bdeb : SUCCESS in 39m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c552e7d2469d4cd899cd62cd6d2895a0 : SUCCESS in 1h 06m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3c30c54130f4aad8f8b32551fce3e02 : SUCCESS in 45m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7528ace1120c4bfd8fa0c6b54eb077db : SUCCESS in 12m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5c6e4640dd514f0facce799074b1a44d : SUCCESS in 5m 19s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fd6ad83e04a14b32938f6659c7581a2a : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/514631cb02aa45ef9cbf5946b6a95b96 : SUCCESS in 5m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/df0709d642d24b3b9326108daed67f6c : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8d98b964af7545ff9da385d34daab9d4 : SUCCESS in 30m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ddd839b2cf7046c9bcf91cfd048c3033 : FAILURE in 7m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9c71aae390204414b5b9de9fefde2a0a : SUCCESS in 1h 09m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0c8b3e1db1044ccf9e620f3a6a314425 : FAILURE in 1h 19m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d238d9e5459045f4a8ceeb82e21cc9ea : SUCCESS in 1h 36m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/041da2b4d54f4ac49926874729c8741c : SUCCESS in 59m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/848043c1e3da4201b004b2752e3090b0 : SUCCESS in 1h 17m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c08a41619c6a4795a91c8fecdbb857f6 : SUCCESS in 42m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f9b475a7267b421cab9713dbc3fac2b9 : SUCCESS in 1h 12m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3349efce46324415b20e253985837fa9 : SUCCESS in 2h 37m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef6c6c4d3583466d992a810b602df51c : SUCCESS in 40m 06s","accounts_in_message":[],"_revision_number":4},{"id":"a0cf2652b8a1e11917152656be0047b05c43f6cf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-27 15:39:43.000000000","message":"Patch Set 4:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"f7f1776fbf0ce7ff1d82eb48c1e75f33031e3eb4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-27 15:39:57.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/843018/4/check/DellEMC_Unity_FC/bf419ad/DellEMC_Unity_FC/874 : FAILURE in 1s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/843018/4/check/DellEMC_Unity_iSCSI/aabac29/DellEMC_Unity_iSCSI/819 : FAILURE in 5s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"ad3cfcd09ff4744fb609abc587c5605df32cb4df","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-05-27 15:57:37.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-18-843018-4/1653665975 : SUCCESS in 18m 00s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"c2dec19ab96565d1bf24f27dbf40021d45a77912","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-27 16:01:37.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/18/843018/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b7c6480/ : TIMED_OUT in 3h 03m 34s\n- pure-devstack-cinder-tempest-fc-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/1bc3a53f5abc49beb1ea45fb4ec551a8 : DISK_FULL in 2h 55m 31s","accounts_in_message":[],"_revision_number":4},{"id":"cb0b133b9451bb4f67a3942ef82a54f4c51f6cc1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-27 16:42:57.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-843018-4/ns5_nfs : SUCCESS in 3h 32m 26s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-843018-4/ns5_iscsi : SUCCESS in 3h 46m 4s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"87e0a0fa118c31d7b85743c537deea2f5b07bf17","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-05-27 17:15:58.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/843018/4/upstream-check/cinder-tempest-plugin-ontap-fcp/e5d3d4d/ : SUCCESS in 1h 08m 31s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/843018/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/f65201a/ : SUCCESS in 1h 28m 36s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/843018/4/upstream-check/cinder-tempest-plugin-ontap-nfs/4de7d2c/ : FAILURE in 1h 27m 13s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"436a654e625dbb399a4934c3bfbe2c39bbb3e200","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-27 19:53:30.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/18/843018/4/check/dsvm-tempest-cisco-zonemanager-job/5df1a0d/ : SUCCESS in 1h 10m 15s","accounts_in_message":[],"_revision_number":4},{"id":"7444115ebde6c4dce642caa485bb518f6e02f531","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-28 13:19:35.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a4c8c508d7ef470094fc94daadbb3d52 : SUCCESS in 1h 36m 26s","accounts_in_message":[],"_revision_number":4},{"id":"0b99d2e2199a0692a0cc30f3ae1b2cd4d258dd66","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-29 02:34:45.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/843018/4/20220529051714/NEC-V-FC : SUCCESS in 1h 48m 39s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/843018/4/20220529051714/NEC-V-ISCSI : SUCCESS in 6h 17m 28s","accounts_in_message":[],"_revision_number":4},{"id":"a6b7e30c8de1d13709d76267e932feb88d6ad0eb","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-06-02 06:31:12.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7ab876a65b518cd9982053ea59d93e7f8d5401b4","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-06-23 09:41:19.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/843018/4/20220623094539/NEC-ISCSI : FAILURE in 6h 21m 15s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/843018/4/20220623094539/NEC-FC : FAILURE in 8h 55m 36s","accounts_in_message":[],"_revision_number":4},{"id":"62b8bec0c6ab77610a2ef9346f91f959347765b6","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2022-07-06 16:48:18.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aebf4be468b55f3ba3880f866b90846105f92a9a","author":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"date":"2023-07-04 13:33:59.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"bc4ddb68194cc50de94b61e4432207524ea31bb0","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-03-25 06:11:00.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c3aded2c0b814d590aca0b4e1cd111d9db7e82dd","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-03-25 06:11:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9328fd0ded5bf75d589493dc3f5e2c0153a632f9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-25 06:12:22.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"fb2e6220ae45682afc9be0dd7d1205d9ecb282ac","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-04-08 13:46:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3cf4ad1a23d12558230fae7dc1bc628b61057a2b","revisions":{"57216a092b067a06e84e9cfab57872c9953ed779":{"kind":"REWORK","_number":1,"created":"2022-05-23 19:25:28.000000000","uploader":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"ref":"refs/changes/18/843018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/843018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/843018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/843018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/843018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/843018/1"}}},"commit":{"parents":[{"commit":"860ce608acbb987580a36611ae8ecfd52882dd6e","subject":"Merge \"Remove reference to removed nfs_*_ratio options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/860ce608acbb987580a36611ae8ecfd52882dd6e"}]}],"author":{"name":"MatheusAndrade777","email":"matheus.handrade15@gmail.com","date":"2022-05-23 18:56:05.000000000","tz":-180},"committer":{"name":"MatheusAndrade777","email":"matheus.handrade15@gmail.com","date":"2022-05-23 19:23:42.000000000","tz":-180},"subject":"Fix NetApp NFS storage migration between backends","message":"Fix NetApp NFS storage migration between backends\n\nMigrating a Cinder volume from a NetApp NFS backend to another\n(different SVM, but same Cluster) was failing in case those\nbackends have same FlexVol name.\n\nThis fix added SVM source and destination parameters to NetApp\nClient to be used as part of API arguments in order to\ndifferentiate the FlexVol volume in the same cluster in case they\nhave the same name.\n\nCloses-Bug: #1969531\nChange-Id: Ia685e398e5c41ae36e1a37530f0a742f79533e9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57216a092b067a06e84e9cfab57872c9953ed779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57216a092b067a06e84e9cfab57872c9953ed779"}]},"branch":"refs/heads/master"},"fbcb41a2785ea1217095b9609fa00b3f8a10b470":{"kind":"REWORK","_number":2,"created":"2022-05-24 12:14:36.000000000","uploader":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"ref":"refs/changes/18/843018/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/843018/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/843018/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/843018/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/843018/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/843018/2"}}},"commit":{"parents":[{"commit":"860ce608acbb987580a36611ae8ecfd52882dd6e","subject":"Merge \"Remove reference to removed nfs_*_ratio options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/860ce608acbb987580a36611ae8ecfd52882dd6e"}]}],"author":{"name":"MatheusAndrade777","email":"matheus.handrade15@gmail.com","date":"2022-05-23 18:56:05.000000000","tz":-180},"committer":{"name":"MatheusAndrade777","email":"matheus.handrade15@gmail.com","date":"2022-05-24 12:12:37.000000000","tz":-180},"subject":"Fix NetApp NFS storage migration between backends","message":"Fix NetApp NFS storage migration between backends\n\nMigrating a Cinder volume from a NetApp NFS backend to another\n(different SVM, but same Cluster) was failing in case those\nbackends have same FlexVol name.\n\nThis fix added SVM source and destination parameters to NetApp\nClient to be used as part of API arguments in order to\ndifferentiate the FlexVol volume in the same cluster in case they\nhave the same name.\n\nCloses-Bug: #1969531\nChange-Id: Ia685e398e5c41ae36e1a37530f0a742f79533e9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fbcb41a2785ea1217095b9609fa00b3f8a10b470"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fbcb41a2785ea1217095b9609fa00b3f8a10b470"}]},"branch":"refs/heads/master"},"48dc3cee37f268bb0b050c7912b91d2ad0d54f0d":{"kind":"REWORK","_number":3,"created":"2022-05-26 14:14:15.000000000","uploader":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"ref":"refs/changes/18/843018/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/843018/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/843018/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/843018/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/843018/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/843018/3"}}},"commit":{"parents":[{"commit":"860ce608acbb987580a36611ae8ecfd52882dd6e","subject":"Merge \"Remove reference to removed nfs_*_ratio options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/860ce608acbb987580a36611ae8ecfd52882dd6e"}]}],"author":{"name":"MatheusAndrade777","email":"matheus.handrade15@gmail.com","date":"2022-05-23 18:56:05.000000000","tz":-180},"committer":{"name":"MatheusAndrade777","email":"matheus.handrade15@gmail.com","date":"2022-05-26 14:12:58.000000000","tz":-180},"subject":"Fix NetApp NFS storage migration between backends","message":"Fix NetApp NFS storage migration between backends\n\nMigrating a Cinder volume from a NetApp NFS backend to another\n(different SVM, but same Cluster) was failing in case those\nbackends have same FlexVol name.\n\nThis fix added SVM source and destination parameters to NetApp\nClient to be used as part of API arguments in order to\ndifferentiate the FlexVol volume in the same cluster in case they\nhave the same name.\n\nCloses-Bug: #1969531\nChange-Id: Ia685e398e5c41ae36e1a37530f0a742f79533e9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/48dc3cee37f268bb0b050c7912b91d2ad0d54f0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/48dc3cee37f268bb0b050c7912b91d2ad0d54f0d"}]},"branch":"refs/heads/master"},"3cf4ad1a23d12558230fae7dc1bc628b61057a2b":{"kind":"REWORK","_number":4,"created":"2022-05-27 12:56:23.000000000","uploader":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"ref":"refs/changes/18/843018/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/843018/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/843018/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/843018/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/843018/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/843018/4"}}},"commit":{"parents":[{"commit":"860ce608acbb987580a36611ae8ecfd52882dd6e","subject":"Merge \"Remove reference to removed nfs_*_ratio options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/860ce608acbb987580a36611ae8ecfd52882dd6e"}]}],"author":{"name":"MatheusAndrade777","email":"matheus.handrade15@gmail.com","date":"2022-05-23 18:56:05.000000000","tz":-180},"committer":{"name":"MatheusAndrade777","email":"matheus.handrade15@gmail.com","date":"2022-05-27 12:55:08.000000000","tz":-180},"subject":"Fix NetApp NFS storage migration between backends","message":"Fix NetApp NFS storage migration between backends\n\nMigrating a Cinder volume from a NetApp NFS backend to another\n(different SVM, but same Cluster) was failing in case those\nbackends have same FlexVol name.\n\nThis fix added SVM source and destination parameters to NetApp\nClient to be used as part of API arguments in order to\ndifferentiate the FlexVol volume in the same cluster in case they\nhave the same name.\n\nCloses-Bug: #1969531\nChange-Id: Ia685e398e5c41ae36e1a37530f0a742f79533e9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3cf4ad1a23d12558230fae7dc1bc628b61057a2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3cf4ad1a23d12558230fae7dc1bc628b61057a2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
