)]}'
{"id":"openstack%2Fcinder~843278","triplet_id":"openstack%2Fcinder~master~I7c36ed6fa46bd11fcd1819b0f042620dba8fa629","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"12988":{"account":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"last_update":"2022-06-22 08:08:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7c36ed6fa46bd11fcd1819b0f042620dba8fa629","subject":"Expose extract_cinder_ids() in cinder.image.image_utils","status":"ABANDONED","created":"2022-05-25 12:59:41.000000000","updated":"2022-06-22 08:08:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"72aaf8c5286c49a2c6b1080365d84a4b53a6f5e0","_number":843278,"virtual_id_number":843278,"owner":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-25 16:20:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-25 13:04:00.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-25 13:27:02.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-05-25 14:32:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-25 14:47:51.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-05-25 16:20:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-25 16:20:33.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-05-25 16:26:32.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-05-25 16:34:42.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-05-25 16:42:33.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-25 17:39:13.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-05-25 18:10:00.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-25 19:13:17.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-25 19:57:35.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-25 20:01:41.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-05-25 22:29:33.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-25 23:30:46.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-05-26 00:18:48.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-05-26 02:24:58.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-05-26 03:18:52.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-05-26 05:06:00.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-05-26 05:34:32.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-05-26 09:01:13.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-05-26 11:18:08.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-05-26 19:58:30.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-05-28 15:09:04.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"e5b18068889903586dbe24eadf2af09b7c9e36ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2022-05-25 12:59:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2107a5039c604e77c871fd4679fa086c1f27aabc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-25 13:04:00.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16072/","accounts_in_message":[],"_revision_number":1},{"id":"1fc6549aedf6c53b0806cf633f9ac06b09877f75","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-05-25 13:06:15.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16072/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"9594bdfc9b26bb40e5d7a6ea82f519facc5fccb6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-25 13:27:02.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1653483590824 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"88023d9435a5860ba1006ed09d5356741c886231","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2022-05-25 13:51:25.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"455f3c62485a6a2e6025ab766ec3ceada1896a90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 14:32:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b40a5816a1742f7bd1ff2afda4a947e : SUCCESS in 21m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d8148c7fffe4add86837f6083f554b4 : SUCCESS in 21m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a39dc5a586cde389e36b7745f65252a23802f447","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-05-25 14:47:51.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"4a707df37f0cd94881092c3098cd0fd833339c0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 16:20:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e9518d3d7a84015b3305aff66aee8ff : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5f72d32e3274c81bf69f6a912e25765 : SUCCESS in 10m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0810f5807abd46d5b4c94f1301568eee : SUCCESS in 11m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0e7db3ad7384882b16e185412d64bb1 : SUCCESS in 11m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ef67ef3b1704926a596efcb4d73497f : SUCCESS in 10m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/03420017230145aa9a9a80d08897f612 : SUCCESS in 1h 02m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b2cae4871a854b78980c191bebbaa880 : SUCCESS in 1h 12m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/270e7cea22da4947af6a8a102274f851 : SUCCESS in 43m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/320038ac139243c18f09c28e063b356c : SUCCESS in 16m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/743b92b09cbf450b98518b1ef6a5846d : SUCCESS in 6m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/65aff5fbcfe84f209c663758cf0a2586 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/60ecbbe40f7544ea8fc5780d1f60265e : SUCCESS in 6m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6ec56e00db96483ab478ed2ff2f959ee : SUCCESS in 7m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cc74d5cc32bb43ebba4aa1bb922dfc8c : SUCCESS in 37m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c41dad45dc17477599d331279a08116d : FAILURE in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/361e4e7f06144cd3978390c35b0f1842 : SUCCESS in 1h 14m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b3ed97b75144434f9a9885a5c18ce9f1 : FAILURE in 1h 35m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b814616215dd46a0b41e9e8c6887ebf3 : SUCCESS in 1h 47m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e17467b9fcc1472cb77ff1b772f4abb6 : SUCCESS in 1h 13m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/86a919ab92b640f9ad8bb255781d5846 : SUCCESS in 1h 26m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d2ab13200db649488da30e644cd67dd1 : SUCCESS in 46m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/34b5f3dac06b47358c585d6e25d8cf72 : SUCCESS in 1h 14m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8de8e416082b48f0bd5a75e47a6f5fc7 : SUCCESS in 2h 23m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2048bf6e47874b2e9c1040a21c50a3c3 : SUCCESS in 56m 43s","accounts_in_message":[],"_revision_number":2},{"id":"ce38ba2dc4ac2600adcd282c40ad3412641e8e91","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-25 16:20:33.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"3b1ad32786c149368dec256324a842912b3d6b6e","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-05-25 16:26:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-solidfire-iscsi finger://managesf.services.openstack.netapp.com:7979/304972e0113a4c808db09a03532484c1 : RETRY_LIMIT in 3m 54s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"eac4d7bf40772e796c079954d738764a357608fe","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-05-25 16:34:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/843278/2/iscsi/ : SUCCESS in 1h 09m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/843278/2/smb/ : SUCCESS in 1h 29m 41s","accounts_in_message":[],"_revision_number":2},{"id":"e9261dff04e9f3387a659b1efcb637c94f1aee72","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-25 16:42:33.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_iSCSI https://elab-os-logsrv.delllabs.net/78/843278/2/check/DellEMC_VNX_iSCSI/8b6547d/DellEMC_VNX_iSCSI/455 : FAILURE in 1s\n- DellEMC_VNX_FC https://elab-os-logsrv.delllabs.net/78/843278/2/check/DellEMC_VNX_FC/d9ba40c/DellEMC_VNX_FC/474 : FAILURE in 10m 59s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"3532ba1e007dd631504bda34fe77c4e02a1088bc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-05-25 17:39:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/843278/2/upstream-check/cinder-tempest-plugin-ontap-fcp/2d00e0f/ : SUCCESS in 1h 17m 20s\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/4a58581b068c417987a0cac096136d20 : RETRY_LIMIT in 4m 06s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/a33a6977454a4699901f67c4ebce9952 : RETRY_LIMIT in 4m 03s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"167b756faa436dcc377dbbda8a62fe910d321bb1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-05-25 17:52:42.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1653486696023 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"6e3372913cbe4e2ab0be7f4c9df37f88b07befc4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-25 18:10:00.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-78-843278-1/ns5_nfs : SUCCESS in 3h 35m 34s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-78-843278-1/ns5_iscsi : SUCCESS in 3h 23m 18s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"5ca69a0bb848f489f126e38f976f36ec509edd0c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-25 19:13:17.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/4c591c5e247c436c987423e4674c2618 : POST_FAILURE in 4h 50m 12s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/78/843278/2/check/devstack-plugin-ceph-dsvm-focal-py3/b753f47/ : TIMED_OUT in 3h 11m 58s","accounts_in_message":[],"_revision_number":2},{"id":"5944316299f44e60436db9648b87cad8a884bb7c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-25 19:57:35.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/78/843278/2/check/DellEMC_PowerFlex/c24a2c7/DellEMC_PowerFlex/796 : SUCCESS in 1h 09m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"73fe76c1a97eae75b269e4391585ff3a07b3ea8a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-25 20:01:41.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/78/843278/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8b321a7/ : TIMED_OUT in 3h 03m 04s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/78/843278/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/5e1b1f8/ : SUCCESS in 1h 13m 34s","accounts_in_message":[],"_revision_number":2},{"id":"06a117a0710fb15c19fea283954d33565af9d1a8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-25 20:53:30.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-78-843278-2/ns5_nfs : SUCCESS in 3h 38m 52s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-78-843278-2/ns5_iscsi : SUCCESS in 3h 41m 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"b55c8721aa979120439ef0d139c73ae3d8bd73d2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-05-25 22:29:33.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/54a75e9a7028417fb5a792261bb62b00 : SUCCESS in 1h 22m 42s","accounts_in_message":[],"_revision_number":2},{"id":"4a45d6cba32a6227ce0d40c7911d09ff47811a3f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-05-25 23:30:46.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/78/843278/2/check/DellEMC_SC_FC/2728e60/DellEMC_SC_FC/643 : SUCCESS in 47m 58s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/78/843278/2/check/DellEMC_SC_iSCSI/f583a1d/DellEMC_SC_iSCSI/664 : SUCCESS in 1h 35m 25s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"a51e51f0c94d973249284670c226c58b00bd97a8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-05-26 00:18:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/78/843278/2/check/iscsi-driver-nimble/c637c8a/c637c8a.html : SUCCESS in 1h 37m 45s","accounts_in_message":[],"_revision_number":2},{"id":"909acf07f5aa357de642b4b22777a181fe954ff9","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-05-26 02:24:58.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/843278-2/HPMSA-iSCSI/ : SUCCESS in 1:08:56\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/843278-2/HPMSA-FC/ : SUCCESS in 1:06:29","accounts_in_message":[],"_revision_number":2},{"id":"23a3bf84306cd29116d1b6f06cf1df65c24b4ab1","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-05-26 03:18:52.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/843278-2/PVME-iSCSI/ : SUCCESS in 1:05:39\n- PVME-FC https://os-logs.seagate.com/dell-ci/843278-2/PVME-FC/ : SUCCESS in 1:02:04","accounts_in_message":[],"_revision_number":2},{"id":"609a2b8a179d40a0195ca2050e84861db8a97718","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-05-26 05:06:00.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/843278-2/Seagate-iSCSI/ : SUCCESS in 1:20:43\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/843278-2/Seagate-FC/ : SUCCESS in 1:19:20","accounts_in_message":[],"_revision_number":2},{"id":"dcf84bad92935279a65a1293277230f4e1f759a6","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-05-26 05:34:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/78/843278/2/20220526115805/NEC-V-FC : SUCCESS in 1h 49m 56s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/78/843278/2/20220526115805/NEC-V-ISCSI : SUCCESS in 2h 36m 24s","accounts_in_message":[],"_revision_number":2},{"id":"93a27a1473b55dbae8caf8a1fd1443050f44e197","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-05-26 09:01:13.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/78/843278/2/check/DellEMC_PowerStore_iSCSI/cab0dfd/DellEMC_PowerStore_iSCSI/541 : FAILURE in 2h 52m 11s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/78/843278/2/check/DellEMC_PowerStore_FC/ebe7a19/DellEMC_PowerStore_FC/566 : FAILURE in 2h 52m 38s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"5b3164d17ba673c90005251ae76f77e27a83d0f8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-26 11:18:08.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/78/843278/2/check/DellEMC_Unity_FC/d7041bd/DellEMC_Unity_FC/841 : FAILURE in 35m 06s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/78/843278/2/check/DellEMC_Unity_iSCSI/cead78d/DellEMC_Unity_iSCSI/806 : ABORTED in 3h 00m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"302d394a33fb65b29f7784101072c99347ed9b8b","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-05-26 19:58:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/843278-2/Lenovo-iSCSI/ : SUCCESS in 1:07:22\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/843278-2/Lenovo-FC/ : SUCCESS in 1:05:33","accounts_in_message":[],"_revision_number":2},{"id":"31267f1597b8ec9d9c2a014e97f376b59613db83","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-05-28 15:09:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/78/843278/2/20220528204553/NEC-FC : SUCCESS in 3h 20m 39s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/78/843278/2/20220528204553/NEC-ISCSI : SUCCESS in 3h 23m 08s","accounts_in_message":[],"_revision_number":2},{"id":"72aaf8c5286c49a2c6b1080365d84a4b53a6f5e0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2022-06-22 08:08:04.000000000","message":"Abandoned\n\nImplemented in a different way in Idbac4dae00f9fa03fb14547a204666721bf67d96","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b256fbdafe1ffb0bc044896c3670fd2c9b0eb9d2","revisions":{"e67d6eb48bbaaf32d18098834aa698bb8b0dee07":{"kind":"REWORK","_number":1,"created":"2022-05-25 12:59:41.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/78/843278/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/78/843278/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/843278/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/843278/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/843278/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/78/843278/1"}}},"commit":{"parents":[{"commit":"07f20139187b1f1f7858f376e7e3edc8260bdae3","subject":"StorPool: create_cloned_volume() improvements.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/07f20139187b1f1f7858f376e7e3edc8260bdae3"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-05-25 11:32:40.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-05-25 11:33:16.000000000","tz":180},"subject":"Expose extract_cinder_ids() in cinder.image.image_utils.","message":"Expose extract_cinder_ids() in cinder.image.image_utils.\n\nSome volume drivers may need to override clone_image() and may have use\nfor this function.\n\nChange-Id: I7c36ed6fa46bd11fcd1819b0f042620dba8fa629\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e67d6eb48bbaaf32d18098834aa698bb8b0dee07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e67d6eb48bbaaf32d18098834aa698bb8b0dee07"}]},"branch":"refs/heads/master"},"b256fbdafe1ffb0bc044896c3670fd2c9b0eb9d2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-25 13:51:25.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/78/843278/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/78/843278/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/843278/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/843278/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/78/843278/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/78/843278/2"}}},"commit":{"parents":[{"commit":"5563b42f8ba435bc45e08399c77ac350156d9aa2","subject":"StorPool: create_cloned_volume() improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5563b42f8ba435bc45e08399c77ac350156d9aa2"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-05-25 11:32:40.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-05-25 13:50:31.000000000","tz":180},"subject":"Expose extract_cinder_ids() in cinder.image.image_utils","message":"Expose extract_cinder_ids() in cinder.image.image_utils\n\nSome volume drivers may need to override clone_image() and may have use\nfor this function.\n\nChange-Id: I7c36ed6fa46bd11fcd1819b0f042620dba8fa629\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b256fbdafe1ffb0bc044896c3670fd2c9b0eb9d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b256fbdafe1ffb0bc044896c3670fd2c9b0eb9d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
