)]}'
{"id":"openstack%2Fcinder~844474","triplet_id":"openstack%2Fcinder~stable%2Fwallaby~I3c0e365f5dc3c32975343f538c6029f02ac7c2b5","project":"openstack/cinder","branch":"stable/wallaby","topic":"bug/1965847-stable/yoga-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-06-28 14:55:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3c0e365f5dc3c32975343f538c6029f02ac7c2b5","subject":"Don\u0027t destroy existing backup by mistake on import","status":"MERGED","created":"2022-06-02 19:03:17.000000000","updated":"2022-06-29 09:56:00.000000000","submitted":"2022-06-28 14:55:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844474-bug/1965847-stable/yoga-stable/xena-stable/wallaby","cherry_pick_of_change":843209,"cherry_pick_of_patch_set":1,"meta_rev_id":"1a5aa54871349c5382581c466ebaab0828f6c97e","_number":844474,"virtual_id_number":844474,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-28 14:55:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"recommended":{"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-29 09:56:00.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},{"value":2,"date":"2022-06-17 15:47:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2022-06-08 11:09:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},{"value":1,"date":"2022-06-17 15:47:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"}],"CC":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-02 19:26:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-02 20:31:33.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-06-02 20:45:01.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-06-02 21:24:05.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-02 21:28:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 22:03:58.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-06-02 22:08:13.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-06-02 22:19:11.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-06-02 22:31:12.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-04 16:54:44.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-06-04 17:15:22.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-06-04 18:54:57.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-06-04 19:05:23.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-06-08 11:09:50.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-06-17 15:47:19.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2022-06-22 01:13:38.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2022-06-28 10:27:19.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"CC"},{"updated":"2022-06-29 09:56:00.000000000","updated_by":{"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},"reviewer":{"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},"state":"REVIEWER"}],"messages":[{"id":"6f52c9e4cc9190f1e277b2770c413677d8f300f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-06-02 19:03:17.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"d524616242f6026d288de026539915163c37f8e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 19:26:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/42756b6c2f6b436483c9f1059ea705d0 : SUCCESS in 16m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3de723e6f7a448cb833d2feb90342f64 : SUCCESS in 21m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5534576cac7c98d6fa6a224cdea5bb143568bab6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-02 20:31:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/844474/1/iscsi/ : SUCCESS in 1h 27m 36s\n- cinder-smb http://cloudbase-ci.com/cinder/844474/1/smb/ : SUCCESS in 1h 04m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e7871ae70a5c7dadcea51dc566f647b6562fe43b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-02 20:45:01.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1654196607902 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"0ae09c9fa178b73c8f7c79b141ddb788c27710ea","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-02 21:24:05.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/74/844474/1/check/tempest-dsvm-full-focal-py3/77587e6/ : SUCCESS in 1h 45m 03s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/74/844474/1/check/devstack-plugin-ceph-dsvm-focal-py3/74e78a3/ : FAILURE in 34m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e55f9ded216eb1982f280d9bdd45524ef2abc802","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 21:28:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ae8addf416a4e44916376c5f7bf53a0 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82448c2ecf4840e09f1d7015ba5a88e6 : SUCCESS in 10m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3bcf6fd3a8b478bb13c3a5220b17ffa : SUCCESS in 9m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2d405428dab408eaa243e90c3129dc7 : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e450ca4856084eb7940b44abad1f0fa8 : SUCCESS in 10m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/8a8dfd3409c940da91eb89c796b8fa0d : FAILURE in 57m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e0f92fe32f0b43fc86b5261bd53a859c : SUCCESS in 1h 18m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0bfcb26aca54bd5b40a4335fc065b00 : SUCCESS in 59m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84c7670a36ec4893bcceb5239ccb065a : SUCCESS in 8m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0eff294cfb4647de968fe11ac95bcccd : SUCCESS in 13m 22s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4aab23f2c7a541439dcddf1fa6c580b8 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/88ad66a7b48c4d9d881a304989cc557c : SUCCESS in 6m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/656fdcafce064af3a6827b7ab6706cc8 : SUCCESS in 5m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/547d9f821acd43f883667430a8cbf7fb : SUCCESS in 47m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d24c8658877d4ff6afe211918e98a6e2 : FAILURE in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e90e26f9de884b3bb865309baf1b961b : SUCCESS in 1h 30m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c76484d5cbae435881b4baec88631dba : SUCCESS in 1h 37m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f76d9f8ddf4b4dfc8edcee1e9f4c4e27 : SUCCESS in 1h 29m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a6ad27c7a74b497ca09464bc0fc8ca57 : FAILURE in 1h 01m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5f30c2a26b914c9cb31f7a243d37ac16 : SUCCESS in 1h 16m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c2d0aa61c02f48c8b6ec0e2deb0a9df8 : SUCCESS in 1h 29m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2bcd7fa5131042d8872bb356424aef88 : SUCCESS in 2h 19m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/929d26cb86b540d195f6c4111604ae01 : SUCCESS in 56m 54s","accounts_in_message":[],"_revision_number":1},{"id":"3349a6897a9faab16f8a1c8ccebd10e8b601bc10","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-06-02 21:34:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fec10d9115a977874579239e0145c831db4957ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 21:58:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1aa6bf24435c4faa9bc2ffeff5cebcc0 : SUCCESS in 17m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2df84a442334f91a44451365c6ed556 : SUCCESS in 22m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1099d3508418a34a63e63b0a276d542f5f1055f2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-02 22:03:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/74/844474/1/check/dsvm-tempest-cisco-zonemanager-job/2810906/ : SUCCESS in 1h 16m 42s","accounts_in_message":[],"_revision_number":1},{"id":"40abd7f8084255a24d6f4d2cf5a64147d2cfee87","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-02 22:08:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/74/844474/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8cbcadb/ : TIMED_OUT in 3h 03m 07s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/74/844474/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0e9e236/ : FAILURE in 1h 19m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e01e679ab7b99e962a209354a090f1af8cd07517","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-02 22:19:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/74/844474/1/check/iscsi-driver-nimble/fa82ff8/fa82ff8.html : FAILURE in 49m 37s","accounts_in_message":[],"_revision_number":1},{"id":"9d4ec243e757e60a76ed2c851b09e9a70cfd71f8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-02 22:31:12.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-74-844474-1/ns5_nfs : SUCCESS in 3h 27m 24s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-74-844474-1/ns5_iscsi : SUCCESS in 3h 18m 27s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"42fe3623d6de39f0da466fa8751a7a6079531671","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-02 23:05:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/844474/1/iscsi/ : SUCCESS in 1h 31m 19s\n- cinder-smb http://cloudbase-ci.com/cinder/844474/1/smb/ : SUCCESS in 1h 20m 00s","accounts_in_message":[],"_revision_number":1},{"id":"cfaee078a65adf7c0fd7e4965666e59a8760dadb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 23:36:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b30584487fba4d818902d092fa628be4 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0e565ee5e1a47778e1196984df9f467 : SUCCESS in 9m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f07fd1b9b11d46319936e77afe813fb5 : SUCCESS in 9m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/157f7898a2154eefb4a268e2c7e2441f : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3eccdd7bb944e83a355de24957da35d : SUCCESS in 11m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/b0ace73e8f144f85b47889e23e796d06 : FAILURE in 57m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a5c4fdb7fce24badb7d3aed18d8eb4be : SUCCESS in 1h 21m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf0b7a30ab6640caac21ac5b9382e33d : SUCCESS in 55m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aab7c03f7b6540438186dca3af9d86bf : SUCCESS in 8m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9899908814c043d6b0e55e0fe9efc737 : SUCCESS in 12m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4970b503177540bdba7e4d96e4f581a0 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d4bfe55e17f642c19388bae2302d89c9 : SUCCESS in 6m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/768aefdbfc1f44beb8d9e56ffeb9e367 : SUCCESS in 5m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2d503376bebb4195afc819b7935fe472 : SUCCESS in 45m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/68d41755b6f84634b7a297d64f31f3ec : FAILURE in 5m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3f0ef74f45ef4491a5929be5d26df7f2 : SUCCESS in 1h 02m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/65edfbed481f4a309c438a1afb8f6568 : SUCCESS in 1h 25m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bf37c5e79d454bbeba825665ea98b46a : SUCCESS in 1h 23m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/df651be9ca2d4cd2ba60f3c8f62ea01b : FAILURE in 47m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b56cf8af2b2a4af3bcc0d084c41e5903 : SUCCESS in 1h 04m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b76123d05833471bbe2f1be2ce5c10db : SUCCESS in 1h 20m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f3f34cbbec044f46b1a58f0b7045ac29 : SUCCESS in 2h 00m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d98b51a90ce647d9a93bbdcc670fe478 : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":1},{"id":"cb14001d797249f4975457160d29c2e956b671c1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-02 23:54:18.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/74/844474/1/check/tempest-dsvm-full-focal-py3/8514d99/ : SUCCESS in 1h 38m 45s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/74/844474/1/check/devstack-plugin-ceph-dsvm-focal-py3/90a79d4/ : SUCCESS in 2h 17m 49s","accounts_in_message":[],"_revision_number":1},{"id":"05c4e35611e10550c6cb20cd2a014ff99de07847","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-06-04 12:43:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e2d572b2b437b10dc0d31981ce1d929f60f5f4ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 13:08:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e68a54628a74ae49f05ad810c9ee248 : SUCCESS in 15m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3dabf1c475c4b89ace33108c520ebe2 : SUCCESS in 22m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a6cdc8fc6d61e96fa97f7fa06981adda3aa9d03","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-04 13:39:57.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/74/844474/1/check/tempest-dsvm-full-focal-py3/5cf4653/ : FAILURE in 40m 27s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/74/844474/1/check/devstack-plugin-ceph-dsvm-focal-py3/f418f8c/ : FAILURE in 33m 36s","accounts_in_message":[],"_revision_number":1},{"id":"2d33053c5f2ea31fe218ecd080932401563d1bd9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-04 14:11:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/844474/1/iscsi/ : SUCCESS in 1h 27m 00s\n- cinder-smb http://cloudbase-ci.com/cinder/844474/1/smb/ : SUCCESS in 1h 06m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9ed62c1d352f458702e0f006d300c4039acebb81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 14:52:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a3d282e0fdb474f9fa7a470b7faf54a : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7541e22cb510433bbbdb02aba8d329f7 : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d27a9dd81ffa478986bd920422344c25 : SUCCESS in 8m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/020f47b860084fd386f1393250dc26f5 : SUCCESS in 13m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7c828db83984a819c98859787754a92 : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/7055164d0f1d4be883d7f0535d05e134 : SUCCESS in 55m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f25b87f26e8340f5a4a0a77dfb971854 : SUCCESS in 1h 01m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fc74dac7e2440c3848d2c2fb53bc173 : SUCCESS in 44m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/783da503321341139cb7989ff1612772 : SUCCESS in 7m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d4cbf2375d084734a9b229bab5c7e1e2 : FAILURE in 13m 11s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/35c384cbe8dd4c228f8ea2649987909e : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d68a43457fa744969e2ed92ec29da881 : SUCCESS in 5m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e8fe7e3cd2174524b4aa677339dd2ae9 : SUCCESS in 5m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/02505821d68f429f9386d730dbefda23 : SUCCESS in 46m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a68a1a70f18432e8c7dae5dd1121e17 : FAILURE in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4a02d2e6474449b4943ba46893bcc64e : FAILURE in 2h 03m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6f7e199ab18f482a9c10b42a87b0c9e6 : SUCCESS in 1h 31m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ce92a60d0e444dc881aee9dfc6e8bae0 : SUCCESS in 1h 11m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b1ed6e52d0e145638337bb9b1e9b19c1 : SUCCESS in 1h 02m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5bdab33476834c9f8a435a0843dcec39 : SUCCESS in 55m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e377968c1aa2449786ce34c4a701e515 : SUCCESS in 1h 03m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e183c44b155541ccba6a696c72a27a03 : SUCCESS in 2h 04m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39f20445bad84416b56aaa1f8aa0dd41 : SUCCESS in 52m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3512e0508b004701305b70df1e58dccded297d56","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-06-04 15:02:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"473b3ee1287ad720aba1147c8c87a58a5901aaba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 15:35:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f128217956b4776b52c542a97958d62 : SUCCESS in 16m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb587a6646ef44ab93d84e68feae6933 : SUCCESS in 19m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"61344d446d1fce2cc658e47126721daffe6e6b01","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-04 15:38:03.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/74/844474/1/check/tempest-dsvm-full-focal-py3/53ee981/ : FAILURE in 30m 17s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/74/844474/1/check/devstack-plugin-ceph-dsvm-focal-py3/259342f/ : FAILURE in 33m 08s","accounts_in_message":[],"_revision_number":1},{"id":"9ec3d53bd04d1727ef2c89c32b31fb63e03b3d68","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-04 16:05:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/844474/1/iscsi/ : SUCCESS in 1h 02m 41s\n- cinder-smb http://cloudbase-ci.com/cinder/844474/1/smb/ : SUCCESS in 1h 01m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3d2553a5cfc4b9b08ff7870bf19e3e0a5d6e90be","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-04 16:44:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/74/844474/1/check/dsvm-tempest-cisco-zonemanager-job/39400eb/ : SUCCESS in 1h 14m 00s","accounts_in_message":[],"_revision_number":1},{"id":"bd5351c75e6f4febcdd6a6a0219c7dafc59d1234","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 16:54:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5804ff5c4c0a4bb78ba406f23758cbd9 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05fc2996e40e485baeb0f0e3368c8118 : SUCCESS in 9m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9686d6c1d1fa421e8fe268c1e8f05189 : SUCCESS in 9m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9f5bfec4908420fb9bd75d3b70078ce : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5912a7fec8e2466a89579e1da98e13c7 : SUCCESS in 15m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/3dfa7879ea574ea2b5fb06bc699977b3 : SUCCESS in 1h 00m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/926c242ccdff47c89b6ac68ef2607745 : SUCCESS in 1h 08m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/215f93fd434f4847abb5b8f2aad0ff89 : SUCCESS in 48m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4397087055314a22b518d15cee9179ce : SUCCESS in 9m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4ae1e8cac07f409d9b5364967e1b86e8 : SUCCESS in 11m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/719e077d0f59404eb94b7ef0ecb5c277 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/66ee4ac19cd14a57977fd97cb6aea8e3 : SUCCESS in 6m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f6b0125f94bd463aafcf15cfe09115c5 : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b87d11df24dd4dd987a2fb92e89a2e4e : SUCCESS in 36m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ad163da5adf646d49f6175c600cb962c : FAILURE in 6m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/831447b2fa3a425799c74b8e8c448b67 : SUCCESS in 1h 15m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6ecc5ed9720548a0a7f1952d0887c6fa : SUCCESS in 1h 23m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/de364152ecf141c8b9888faac7ceda5d : SUCCESS in 1h 26m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f76045e2d3fe46d292c9ae1f9c93379d : SUCCESS in 1h 18m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/43271ba97ba8438ebd2980cfb57859c4 : SUCCESS in 1h 15m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3471ab1dcdfa444cbd66449ed9c1d50f : SUCCESS in 1h 11m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/845b670cd9904b2e8721c661eba6a0e8 : SUCCESS in 1h 45m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ad7c66861b14dbeaa63a2ddd30502da : SUCCESS in 51m 37s","accounts_in_message":[],"_revision_number":1},{"id":"437f04379efae8a5a08f193ced5b93ba086c7052","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-06-04 16:54:44.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"0df1c064e0de1dba8a3a20bfe52354f29dae1ee9","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-06-04 17:15:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-74-844474-1/1654361673 : SUCCESS in 20m 47s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":1},{"id":"d1f1b8be59ac17238131eb4c91c4ca80387f9141","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-06-04 18:54:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/844474/1/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/82f7c43/ : POST_FAILURE in 1h 52m 15s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"1753f5d56293a6caea26f5f0b258928edb51683b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-04 19:05:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/844474/1/upstream-check/cinder-tempest-plugin-ontap-fcp/4a62a07/ : POST_FAILURE in 57m 37s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/844474/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/3507165/ : POST_FAILURE in 2h 08m 54s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/844474/1/upstream-check/cinder-tempest-plugin-ontap-nfs/3d3931e/ : POST_FAILURE in 2h 08m 25s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"25d796a38fd75ccfa8880bf4d170986a00f3bc15","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-06-08 11:09:50.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f9262074c5aed9e924b9af762339e565cb21f9a","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-06-17 15:47:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77710ffbf15eb2c1768ed33f3b9803a98c5c8a29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 15:48:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4592c19b14311b0fc345dca96f665f4f353f1493","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 20:38:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/405cd1816126419b8636e0eaa15de19a : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96b6b62457b5407da6b2749a861f1322 : SUCCESS in 9m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad45dc1928af46d699f7cbc86c615f8a : SUCCESS in 10m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4507f40807be44ecb76b02edd0e032e5 : SUCCESS in 11m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/1c74421b6b0d46dbaedf0cc6342a3d71 : SUCCESS in 1h 08m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3220ed0d550e4b9dbb0c151c90449089 : SUCCESS in 1h 21m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/100fbe852a79481e878c7bd7fb1be3ce : SUCCESS in 1h 01m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12c0fbf03b344b6c87c8189cbb7f7745 : SUCCESS in 8m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/048290d67c6549dfb445bc78c2c901cd : SUCCESS in 1h 22m 44s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9f1e4fef55d949b9b5d936d5c5615243 : FAILURE in 1h 34m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61380e3f4fff414caea58660325c3c40 : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":1},{"id":"4c1c35e0f135bf3e0fa4c66f2ac9c0d6817e3d83","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-06-21 23:05:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d16ce8fdbb646a053fec393f9adf9dd094ffade8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-21 23:44:03.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/6ca1d5aeaf5245fa9b44e2c93019156d : POST_FAILURE in 35m 54s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/d143ec163ac3452d9a384f2f4bf54298 : POST_FAILURE in 32m 55s","accounts_in_message":[],"_revision_number":1},{"id":"879938aec70ab6d288d60204f6e68b01ab8d004c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 00:08:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f2f4814ef814a1eb81cf0c600d6dd7d : SUCCESS in 17m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3c394de237c548c9b35ca590273a6165 : SUCCESS in 23m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0be3c0d5db4c86d08c9f4cc93dff6e863ea17764","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-22 00:19:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/844474/1/iscsi/ : SUCCESS in 1h 07m 11s\n- cinder-smb http://cloudbase-ci.com/cinder/844474/1/smb/ : SUCCESS in 1h 13m 53s","accounts_in_message":[],"_revision_number":1},{"id":"17f75244cea0d8e4dad37c13a9af545796eb5252","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-06-22 01:13:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi finger://vm-css-bionic-2/36153aa956e740d0bc72d46ba88e2826 : POST_FAILURE in 2h 07m 17s\n- storwize-tempest-dsvm-full-fc finger://vm-css-bionic-2/cabe06cac2c347bab5b278fe37efdc1c : POST_FAILURE in 2h 07m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f5d46650e058a8a22ced936047f8b446acf0ce8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 01:34:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45843608849346ec84f666d8cd466065 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86bfc99ddfb942ec9e5bb7a8190e91eb : SUCCESS in 9m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b69975d60d744460a1c4144d8a066a98 : SUCCESS in 10m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccfb1223c0204850b5aa0b86eba1053a : SUCCESS in 12m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a30653fb642f4215ab97823c7fd1c540 : SUCCESS in 11m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/a1a7c1f187db4a72b90a00d8c36cb548 : SUCCESS in 59m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fa2f2acea2ac42e3b888e9e90a02a272 : SUCCESS in 1h 02m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a51ed436723430aa66ea78ea3bc7907 : SUCCESS in 56m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa65d4a038b1499f99aea6df0c8db2b2 : SUCCESS in 8m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/af23b259b4704429bfb9c548c2832f82 : SUCCESS in 10m 52s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3533e98f672d45a7815db8e9763abbd5 : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/767c6d2215f1402184430be5a92816de : SUCCESS in 6m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b8163260d16243279b8c7abd8b89b0f3 : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4a693799d38f4b3e89f7f905ddb93303 : SUCCESS in 51m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1899fa96e73840e18e63519c2066ee35 : FAILURE in 7m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2d55017790874e2e9cc9af94799b8136 : FAILURE in 1h 13m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2234003fb91640468b570b4e95cba376 : FAILURE in 1h 39m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0d901e8a27194a8b8da6996967ebdb61 : FAILURE in 1h 30m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c3b1c89cb8014e9a906a3a1983aeb22c : SUCCESS in 1h 08m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7632723b19014339a051c1749ba48392 : SUCCESS in 1h 20m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9c69266f394347d69ba6dcf9c5bcc7fb : FAILURE in 1h 06m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/564326b6b85b40e3b883e40e9cedca46 : SUCCESS in 2h 23m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4b496fc46e3403f9501b3263d795e15 : SUCCESS in 53m 55s","accounts_in_message":[],"_revision_number":1},{"id":"6ea17c6453082ab511513359b845e9d2052fe387","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-22 02:15:10.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/74/844474/1/check/dsvm-tempest-cisco-zonemanager-job/e3b40aa/ : SUCCESS in 1h 17m 23s","accounts_in_message":[],"_revision_number":1},{"id":"af296557b03fe48acd9a51036d531d55e8b305e5","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-06-28 10:27:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc1eb3af93126a164806a085236471d7d2068fb2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 10:51:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/91b29e1f6c8e457b9c405b61b22a621e : SUCCESS in 17m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dde5c651d50c4f16936ab707fcf0168e : SUCCESS in 21m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef3f6e29fd3758f71739800c522e01843dcab5ea","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-28 11:16:43.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/74/844474/1/check/tempest-dsvm-full-focal-py3/32fae00/ : FAILURE in 41m 30s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/74/844474/1/check/devstack-plugin-ceph-dsvm-focal-py3/8b6a40b/ : FAILURE in 45m 44s","accounts_in_message":[],"_revision_number":1},{"id":"aab35225ce342c0ea713a1502c4b2f1af1c6db91","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-28 11:42:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/844474/1/iscsi/ : SUCCESS in 1h 14m 48s\n- cinder-smb http://cloudbase-ci.com/cinder/844474/1/smb/ : SUCCESS in 1h 04m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c73a5e8c8eb76ff6343da7143509691868aac669","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-28 12:30:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/74/844474/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/096ffb7/ : SUCCESS in 2h 02m 12s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/74/844474/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/2793482/ : SUCCESS in 1h 55m 32s\n- pure-devstack-cinder-tempest-nvme-fc-aio http://openstack-logs.purestorage.com/74/844474/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-fc-aio/5b53f21/ : FAILURE in 33m 05s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/74/844474/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/785e96a/ : FAILURE in 32m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cccf0708c71bad6b9f4363d13760d6559fb56a76","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-28 13:03:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/74/844474/1/check/dsvm-tempest-cisco-zonemanager-job/fcc1aea/ : FAILURE in 1h 04m 38s","accounts_in_message":[],"_revision_number":1},{"id":"fb3daa4978f106962cf6ebfaff45ab96558a0ffe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 13:05:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8c045f571f848c0b6bdf50754849660 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7815111475b46de9461d9a966ab1a18 : SUCCESS in 9m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38cb3e3dce6945f0bb8b8f1b0feede2e : SUCCESS in 9m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33658eba375b4f56a59335eb811e4d15 : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d8693a8e0d94aa58c0b4123ecc37ed5 : SUCCESS in 13m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/63feef1b562343bd93ea0ef0132fe6fc : SUCCESS in 59m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5fe568daabb2466db35def65911a42ed : SUCCESS in 1h 22m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33beaf6f27654adb83f5698c2983472e : SUCCESS in 54m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdeff656894f4d2388730f3c23b9ae91 : SUCCESS in 8m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/33a87723f79b4dd0bb5ef5f0c8ba7b62 : SUCCESS in 12m 40s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9a0a234637b94b16b2b8648c67e9fcfa : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/328a723bbb01402b9c96b00ac0e12718 : SUCCESS in 6m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8ff506bd66b94aa3af65c90b5a41ae33 : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a52a471e31f5427c8e0759a4b6588f4a : SUCCESS in 51m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/86e9e079844344b2b7126ab036e960db : FAILURE in 5m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4f25dd6b6ee240a798f695bcf6ec6c3c : SUCCESS in 1h 12m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bdfbeadf2fcb4fb58b051926134b9348 : FAILURE in 1h 26m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7fd2a925376844ebb455dd3fe061ef45 : SUCCESS in 2h 03m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2b94744f044a48d6abed0477be3af013 : SUCCESS in 1h 07m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/af5a5adb7ff14fce804e963b27620f1a : SUCCESS in 1h 25m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c0287c5058d5421abeacd10952057690 : FAILURE in 1h 11m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e956ad47ea3b41028caac4b3653fd4c5 : SUCCESS in 2h 32m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9818b8aaec464df9802492982d3f3d44 : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b338d93de59eedc42255b23ee80c700d7de34238","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 13:06:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6db07e50789f7410afe9657cd99dac1c11b29c5","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-06-28 13:25:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-74-844474-1/1656421537 : SUCCESS in 19m 54s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":1},{"id":"07a832bf49532387746267c8748856fd7ad7ae0a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-28 14:19:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/844474/1/upstream-check/cinder-tempest-plugin-ontap-fcp/46d9df2/ : POST_FAILURE in 1h 08m 50s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/844474/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/c8942a9/ : POST_FAILURE in 1h 13m 14s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/844474/1/upstream-check/cinder-tempest-plugin-ontap-nfs/f231705/ : POST_FAILURE in 59m 16s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c9dd50efe1ccc4e005214bf5465630455aef6398","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 14:55:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d36a7285f9a9436cb014453fe00aad2c : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/520f21a831134bac84763c1b9f9c5962 : SUCCESS in 9m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6663304af21344e49d44c59e473a1614 : SUCCESS in 10m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bd8c8c55b104f2db38ea874bec7ac6d : SUCCESS in 12m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/f260d7d4efaa42568ffff72f6ec4a147 : SUCCESS in 58m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a57c574b1584ce59b0e9b9ad8643c74 : SUCCESS in 1h 13m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee22d6f77f9d4339a5048e754ca9c9d1 : SUCCESS in 49m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0b2ba173a0245b095bf017748d00b0c : SUCCESS in 9m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e6267b801dac4a5eaba1e8bd32cdf468 : SUCCESS in 1h 21m 45s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e6284ba6fde646f7a78f41276af374c8 : SUCCESS in 1h 42m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab7e05b6ebd745dfbe79ed8fa875e7d0 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d584a4e8e639bf1c4b923dfa314ae018e740092f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 14:55:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cf93f99a50b5bb093caad71a15c7eff05d6a3f11","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 14:57:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef870377d82740d5bd88b0926a7e8f89 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b785020dd0b4f5ea757649ad171257e : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1a5aa54871349c5382581c466ebaab0828f6c97e","author":{"_account_id":34541,"name":"Manish","email":"manish.laddha@ibm.com","username":"manishladdha"},"date":"2022-06-29 09:56:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"741d0ce546d3221329751fd024cade09088d6422","revisions":{"741d0ce546d3221329751fd024cade09088d6422":{"kind":"REWORK","_number":1,"created":"2022-06-02 19:03:17.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/74/844474/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/74/844474/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/844474/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/844474/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/74/844474/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/74/844474/1"}}},"commit":{"parents":[{"commit":"553264535dc34f676ff4db3c77e1869c1fb3c6b7","subject":"Prevent temporary volume from being deleted accidentally","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/553264535dc34f676ff4db3c77e1869c1fb3c6b7"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-04-26 22:57:36.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-06-02 19:03:17.000000000","tz":0},"subject":"Don\u0027t destroy existing backup by mistake on import","message":"Don\u0027t destroy existing backup by mistake on import\n\nWe\u0027re supposed to fail an import request for an already existing\nbackup, but the current code confuses an existing backup record with\none we created for the import, and during quota rollback, will\ndestroy either one.  Simplify the logic so that we check whether\nthere\u0027s an already existing backup first and bail without even\nbothering to make an unnecessary reservation, hence leaving us nothing\nto roll back or delete by mistake.\n\nAlso revise and add tests.\n\nCloses-bug: #1965847\nChange-Id: I3c0e365f5dc3c32975343f538c6029f02ac7c2b5\n(cherry picked from commit 6386cbb0a0b6cbf05ce625c5b1b680f4259bde2e)\n(cherry picked from commit 753c0224631e218c860e48722520e9ad384ca2f4)\n(cherry picked from commit 332aa33404f53b39392df3fc37dddb85b5a2890c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/741d0ce546d3221329751fd024cade09088d6422"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/741d0ce546d3221329751fd024cade09088d6422"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
