)]}'
{"id":"openstack%2Fcinder~844961","triplet_id":"openstack%2Fcinder~master~I3be92fcd4ac1f86230adf7c2998e2f0100a2ec33","project":"openstack/cinder","branch":"master","topic":"remove-legacyfacade","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-09-13 15:43:15.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-09-13 12:10:27.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I3be92fcd4ac1f86230adf7c2998e2f0100a2ec33","subject":"db: Remove unnecessary engine facade decorator","status":"MERGED","created":"2022-06-07 12:23:04.000000000","updated":"2022-09-16 11:24:58.000000000","submitted":"2022-09-13 15:43:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"844961-remove-legacyfacade","meta_rev_id":"3211495ffba8c7725db94ce9bc2cfbbebe7981bc","_number":844961,"virtual_id_number":844961,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-13 15:43:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-31 08:56:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":1,"date":"2022-08-31 13:01:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":2,"date":"2022-09-13 12:10:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":1,"date":"2022-09-13 12:10:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-07 12:49:00.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-06-07 12:54:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-07 13:08:54.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-07 13:45:50.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-06-07 13:56:46.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-07 14:13:33.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-06-07 15:28:31.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-06-07 15:39:39.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-07 15:49:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-07 16:02:47.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-07 17:02:49.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-06-07 23:29:05.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-06-13 05:31:18.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-23 22:58:08.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-07-01 18:05:37.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-07-01 19:00:33.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-07-01 19:43:06.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-01 19:46:44.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-07-02 18:55:20.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-06 13:59:35.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2022-07-10 14:30:41.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-31 08:56:46.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-08-31 11:01:22.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-31 11:35:09.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-08-31 13:01:57.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-08-31 23:45:45.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-09-01 14:48:42.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-09-01 23:51:15.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-09-01 23:51:30.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-09-02 08:17:53.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-09-09 10:48:26.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-13 09:34:26.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-09-13 13:40:16.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2022-09-14 13:17:02.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"state":"CC"}],"messages":[{"id":"562a24fd2c6b80eb7572cc70e206258b9ee281f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-06-07 12:23:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e96f905c67de83bebdde0d6c879dd5811677f65","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-07 12:49:00.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1654604594183 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"98a31953f88b4daea1bf43be0acaf0d8f135888c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 12:54:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a2650f315aa4aedaa76b9f3f30c6909 : FAILURE in 27m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0375c48464244a980f58fd3601e3bf6 : SUCCESS in 28m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ae7d9bf23e7adfd78fc06f193c702c663aed33b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-07 13:08:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/438cfdd19d5e48b8bb453714bccd8c6b : SUCCESS in 44m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c7ac5b87515f0b15e1889b596eaa075e4b1f8011","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-07 13:45:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/844961/1/iscsi/ : SUCCESS in 1h 21m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/844961/1/smb/ : SUCCESS in 1h 05m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0d34cf1d0be7a0b382c2455d09bc0b77b1933f39","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-06-07 13:56:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/844961 : SUCCESS in 1h 30m 55s","accounts_in_message":[],"_revision_number":1},{"id":"9986669014d30f056a3bc2288ad56728f945e6ca","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-07 14:13:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/61/844961/1/check/dsvm-tempest-cisco-zonemanager-job/08d9930/ : SUCCESS in 1h 18m 31s","accounts_in_message":[],"_revision_number":1},{"id":"68e51ca3f7ca3cfabb7fcb9f98e788a8a97574cc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-07 15:28:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/61/844961/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/88c3953/ : TIMED_OUT in 3h 03m 37s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/61/844961/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a9c163e/ : FAILURE in 1h 21m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ee4e90cc09ee4a272092f9ad585e0288413012cc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-07 15:39:39.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/61/844961/1/check/tempest-dsvm-full-focal-py3/07ead8b/ : SUCCESS in 2h 26m 35s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/61/844961/1/check/devstack-plugin-ceph-dsvm-focal-py3/8000554/ : TIMED_OUT in 3h 11m 41s","accounts_in_message":[],"_revision_number":1},{"id":"80b8d2f4c0ddeaa663a4d2d87b832f89bd3a2750","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 15:49:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73359aa92abc4edcb3444c862f906ddc : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/146728a571354b0883371fc0511d526d : SUCCESS in 12m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9c65ff9c3544ec3a186537de33c0ae4 : SUCCESS in 13m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22a8e48e63fa45aa9cb45725550e63e2 : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e80f3c779b2c46c1a0ece45a73dec20b : SUCCESS in 13m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/3f15b6ffca3447b1a19aec2f5c1af20c : SUCCESS in 50m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e1202ee1979548e79704e62f7b337819 : SUCCESS in 1h 06m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ca88995a14a41aebc82d6dd7b8ba7ef : SUCCESS in 44m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7491457347314281bfc8a69f5ab2f3a2 : SUCCESS in 17m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bb102cd7faa44f28b30e0ebec6f0dc37 : SUCCESS in 4m 48s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/05d0239ca64447d198117c46b3f23404 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/802f28b245334e0590a26a7a5a99607a : SUCCESS in 6m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1bf8bf54a55143108ac9858696411f64 : SUCCESS in 6m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e62a2af4c18c4e88a9f024aef06da0eb : SUCCESS in 49m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7363e69b5a8345f2ae8adfeaa2941195 : FAILURE in 7m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/28f5853f0a6e4cdf928f5c097eb491bd : SUCCESS in 1h 07m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/54e5524f7b6d4e08a39648e685582adf : TIMED_OUT in 3h 09m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0c57d0f09e854874b4be4782fd0289a8 : FAILURE in 2h 01m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bc2c896b68bf438fabc841aeea9f97f8 : SUCCESS in 1h 04m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4731848c4b5c42e39d22ddbec35e2b2a : SUCCESS in 1h 21m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c6fd04d25a2d40d9accd957575001e30 : SUCCESS in 39m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/134436ea8b084c7e877bdfaac7d34356 : SUCCESS in 1h 18m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/413938b0778140e7af0d0c5af92c6cdf : SUCCESS in 2h 35m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22a59dcf34d947648afe83dd73d194b9 : SUCCESS in 45m 37s","accounts_in_message":[],"_revision_number":1},{"id":"8be66c8a6e894b07254cc32692c35d132050f082","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-07 16:02:47.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-61-844961-1/ns5_nfs : SUCCESS in 3h 35m 53s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-61-844961-1/ns5_iscsi : SUCCESS in 3h 39m 13s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"9baf8a3cdf0ef84091a1c187c6b7d88640794bc4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-07 17:02:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/61/844961/1/check/iscsi-driver-nimble/0a6dfc9/0a6dfc9.html : FAILURE in 1h 37m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e3fe6f96dd0d08281c45488742903af6dabbf4b1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-07 23:29:05.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/61/844961/1/check/DellEMC_Unity_FC/4efbb58/DellEMC_Unity_FC/934 : SUCCESS in 1h 06m 41s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/61/844961/1/check/DellEMC_Unity_iSCSI/0460d56/DellEMC_Unity_iSCSI/878 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d7da269d0898b4126aa63f18de175bf8cb32e1af","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-13 05:31:18.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"f245ad81ea45a8bc3ae10bd46ae5ab482991a641","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-06-23 22:58:08.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/61/844961/1/20220624012811/NEC-V-FC : FAILURE in 5h 39m 16s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/61/844961/1/20220624012811/NEC-V-ISCSI : FAILURE in 6h 29m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d212acde33cf25941e251f0dea466905fc549137","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-01 15:38:35.000000000","message":"Topic remove-legacy-facade removed","accounts_in_message":[],"_revision_number":1},{"id":"7c2f6ff004c83da9d1d725369daff7783caec8cc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-01 15:38:38.000000000","message":"Topic set to remove-legacyfacade","accounts_in_message":[],"_revision_number":1},{"id":"f3171190bac90eb6f12d5e500a2dc3339bee3c6f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-01 15:38:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7aa88159257e8b5c8911b19bfdbf5028430710b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 16:09:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9979aaea30bb4ba48eee9da7f4a153d8 : SUCCESS in 17m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c65dcc31696a41448b5baa00ffa57d20 : SUCCESS in 21m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed8e4a43c3e7876d58b09ff213b80247d0735f8c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-01 17:01:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/844961/1/iscsi/ : SUCCESS in 1h 22m 04s\n- cinder-smb http://cloudbase-ci.com/cinder/844961/1/smb/ : SUCCESS in 1h 08m 00s","accounts_in_message":[],"_revision_number":1},{"id":"9c1baa8d4815919c46e5854d96e1802bb02ea8a8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-01 17:20:20.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/61/844961/1/check/tempest-dsvm-full-focal-py3/02c01c6/ : SUCCESS in 1h 16m 40s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/61/844961/1/check/devstack-plugin-ceph-dsvm-focal-py3/54e2b3c/ : SUCCESS in 1h 39m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3805a93da4337aacbf61ce0e67441ac067099b3a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-01 17:22:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : RETRY_LIMIT in 30m 23s","accounts_in_message":[],"_revision_number":1},{"id":"da901cd5d2e70b848ed310829b1ad99ac493a554","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 17:48:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c47080aac4d5427c95dcd61c273f4e02 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/274e2af4e8d5443a9ad3a2142619ffeb : SUCCESS in 11m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0745f3326c654e81ae49166f1e96c556 : SUCCESS in 8m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbd7d137b1a54d77a6b335400fc57d15 : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03fb97da5c2b44bfb7e84f978f7dcc0a : SUCCESS in 10m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/c6ad894fcc374e0f96837d626ce19b10 : SUCCESS in 51m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/83ff8b7354894ef2b68a0ad128a67ac5 : SUCCESS in 1h 05m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93c2275fb6da45628e852d64e6c6df51 : SUCCESS in 41m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0f4f6933dc0f4750b0b960e3cb52dd4d : SUCCESS in 13m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f20725356bbe4766ba3b1ff5a05c1216 : SUCCESS in 4m 49s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/daa480f6fc34466aadcd8388705775a0 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e50e260546e346de8d2a4f88de5a5357 : SUCCESS in 5m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/768e2580485d46a9a550c47bbc1aa509 : SUCCESS in 7m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a8108dcccf394e97b1ba2cc5f73a2e45 : FAILURE in 22m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c363623fc582487a9997e4b735cce68b : FAILURE in 6m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/04cd92deb2084017a1f2f734f4ada9b4 : SUCCESS in 1h 16m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8063a810b9f3450198d6369595418386 : FAILURE in 1h 27m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b3766831bc2c430abe0350203aa91242 : SUCCESS in 1h 42m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ae5677feff054825a6e5dd67b536640e : SUCCESS in 1h 01m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/16ea6abd4f004862ae1db789b1a7c0a5 : SUCCESS in 1h 07m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fb8951dd654240eba36547ecc3ba2f63 : FAILURE in 35m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f5d5a20cdc5c42e3ac7501f621c140ac : SUCCESS in 1h 20m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2c5d7de00d3a469585f6e09ab262748b : SUCCESS in 2h 04m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a8db0cb7921487fbb2d7b79c34a5e0a : SUCCESS in 50m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a0edd1fcd13efea6bffb3bd0991b9c141558762d","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-07-01 18:05:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-61-844961-1/1656697708 : SUCCESS in 17m 08s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":1},{"id":"96e9cb8d3252aae388999ff03d04491b2c57cb03","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-01 19:00:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/61/844961/1/upstream-check/cinder-tempest-plugin-ontap-fcp/2f001c7/ : SUCCESS in 1h 11m 06s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/61/844961/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/e850b30/ : SUCCESS in 42m 54s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/61/844961/1/upstream-check/cinder-tempest-plugin-ontap-nfs/37023ff/ : SUCCESS in 41m 12s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d3c624672d924182259f27e9a2f79fde97f0b514","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-01 19:43:06.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/61/844961/1/check/DellEMC_PowerFlex/cc0e555/DellEMC_PowerFlex/939 : FAILURE in 1h 54m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"2d31795693c35c804b46c6850490140c4ac823b1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-01 19:44:04.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/61/844961/1/check/DellEMC_Unity_FC/e30b11d/DellEMC_Unity_FC/1069 : SUCCESS in 1h 13m 59s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/61/844961/1/check/DellEMC_Unity_iSCSI/b12d32e/DellEMC_Unity_iSCSI/1003 : SUCCESS in 1h 54m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"7ae2380dbe00798658cf73e0fc54279355898011","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-07-01 19:46:44.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/61/844961/1/check/DellEMC_SC_FC/bb8603b/DellEMC_SC_FC/779 : SUCCESS in 47m 00s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/61/844961/1/check/DellEMC_SC_iSCSI/48f2f1b/DellEMC_SC_iSCSI/801 : SUCCESS in 1h 57m 59s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"6fb050acec3c1b606afdf9050b8a9a2ac2ba46f2","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-07-02 18:55:20.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/61/844961/1/check/DellEMC_PowerMAX_FC/5cc534c/DellEMC_PowerMAX_FC/663 : SUCCESS in 4h 27m 54s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/61/844961/1/check/DellEMC_PowerMAX_iSCSI/b734210/DellEMC_PowerMAX_iSCSI/649 : SUCCESS in 4h 05m 22s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"b07ad0d612f64278477121cdd1b5ba0408f30b3f","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2022-07-06 13:59:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aa5ce9ea939431fcc55fcd9589faf1db628b58db","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-07-10 14:30:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/61/844961/1/20220710195858/NEC-ISCSI : SUCCESS in 3h 30m 55s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/61/844961/1/20220710195858/NEC-FC : SUCCESS in 3h 31m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e9fb9005e9f48295631ace109bcc06d12b5548b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-08-31 08:56:42.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"136ba1c06dfe79ba369502035d72ff7255b11c79","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-31 08:56:46.000000000","message":"Patch Set 2:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"d2f7930dab67e48961e434230ac17aebdf14534e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 09:22:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/661db71fe546444b9528eb0224827c5e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5db1c1f394554ca89187470029a72e92 : SUCCESS in 20m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df5fd1063a1c41d0a41ccbab14ac2c9b : SUCCESS in 17m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dbc847ff9a9b837c4c592d2f65d1e8c5edf5e7a8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-31 10:10:00.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/844961/2/iscsi/ : SUCCESS in 1h 12m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/844961/2/smb/ : SUCCESS in 1h 10m 11s","accounts_in_message":[],"_revision_number":2},{"id":"e79cfc0f4b102be0b1eb7f73b4c8f219f54ba079","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-31 10:25:27.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/844961 : SUCCESS in 1h 26m 33s","accounts_in_message":[],"_revision_number":2},{"id":"64e138f8e35cb2dc59527131c09e357a3d289986","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-31 10:28:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/844961 : SUCCESS in 1h 29m 13s","accounts_in_message":[],"_revision_number":2},{"id":"983db285a118dc688f1f7a0e04275e0f1b859e1b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-31 10:34:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/61/844961/2/check/iscsi-driver-alletra-6k/a9a63a2/a9a63a2.html : FAILURE in 1h 35m 43s","accounts_in_message":[],"_revision_number":2},{"id":"a6fb06832e31628c407b85a2bc5beb317edf29a8","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-31 11:01:22.000000000","message":"Patch Set 2:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/61/844961/2/check/fujitsu-eternus-dx-iscsi/517fe64 : SUCCESS in 1h 58m 48s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/61/844961/2/check/fujitsu-eternus-dx-fc/073e1fe : SUCCESS in 2h 03m 53s","accounts_in_message":[],"_revision_number":2},{"id":"70708346d2cf75c76c304fe1b0efbd4d0e27de41","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-31 11:13:31.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/61/844961/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/9639a79/ : SUCCESS in 2h 13m 36s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/61/844961/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/296639c/ : SUCCESS in 2h 09m 19s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/61/844961/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/45bc41f/ : SUCCESS in 1h 52m 32s","accounts_in_message":[],"_revision_number":2},{"id":"c6d2602a2b0d5ace68c6185b1d5b09752f408afa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 11:27:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e78c4fdbeea849ab9d5c7d8959e30e2a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de34971d17cb4573840692636dedb412 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8eb1dc2204ba4af3a07325f28ef16c1a : SUCCESS in 9m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7987e20a056a4e33856df70ce521d16f : SUCCESS in 10m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78a1f47876644ee0809ad079f7d27d64 : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb97d3db93c7454a92edd4a56a2666ee : SUCCESS in 11m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/8e8643e216374beb95f0bc14277476e0 : SUCCESS in 53m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4ab966bb7ad140b9ae1923a485a1ed49 : SUCCESS in 1h 22m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ab99167a6684d40b9f95df5347c4535 : SUCCESS in 1h 04m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8482c457729c4c229db8b750990f0e0d : SUCCESS in 14m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/84fd4d2f3493467f87a8285c54fc4f49 : SUCCESS in 4m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5cbf617a51fe40ccac5ca85522d4010b : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3a07786afa794531bbb1bafe18765a57 : SUCCESS in 6m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/64472303f34b48bfbf1978f91d8e6e89 : SUCCESS in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0650acc350b942b095e1a3af22844674 : SUCCESS in 37m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/62e073cc95b54d5faacbb9905eb32db6 : FAILURE in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2979a942d4ee4088a13c4cc8bde47e40 : SUCCESS in 1h 15m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4d29e1f2eef74925a7312e9fa3feedd2 : FAILURE in 1h 47m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/58a53a111d444e698193cf42209293e5 : SUCCESS in 1h 31m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9e23d6dbd56d492b94fa2a19874085bf : SUCCESS in 1h 13m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4b71b239287a4c87ac1c0a96155e5d15 : SUCCESS in 1h 26m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5bd1c1805f0744cf8d9bfea7054afb7a : SUCCESS in 36m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c6f7d769696b456283889b61f89f6c27 : SUCCESS in 1h 49m 16s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ecb0a20bd4774d3f8fc20e2aca6d1964 : SUCCESS in 2h 24m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/564890c31ede44d2b6f6857dfc5b5721 : SUCCESS in 50m 51s","accounts_in_message":[],"_revision_number":2},{"id":"75e6476ae20e65db5cad8c328b02d2b6fb500652","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-31 11:35:09.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_08_2022_11_27_22\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"e24ee40c82e27823a57ec246a31694468cd67a87","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-31 12:28:43.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-61-844961-2/ns5_nfs : SUCCESS in 3h 31m 29s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-61-844961-2/ns5_iscsi : SUCCESS in 3h 25m 26s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"1aa509007ea5c81c84c75b78435a0709600e9f4c","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-08-31 13:01:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0b21e42c46f1c883519d865e403d72beae9dfb17","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-31 14:47:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/61/844961/2/upstream-check/cinder-tempest-plugin-ontap-fcp/e0654d2/ : FAILURE in 1h 31m 17s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/61/844961/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/591db81/ : SUCCESS in 45m 54s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/61/844961/2/upstream-check/cinder-tempest-plugin-ontap-nfs/12c4c5b/ : SUCCESS in 1h 06m 47s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/61/844961/2/upstream-check/cinder-tempest-plugin-solidfire-iscsi/6ad7153/ : SUCCESS in 1h 34m 34s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"e61d50c6eec95654d6877860757e6b4eccb64227","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-31 21:25:48.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/61/844961/2/check/DellEMC_PowerFlex/245063a/DellEMC_PowerFlex/1261 : ABORTED in 2h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"3a5d2302bc70b0aca4ea4724e83e21c517b2815f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-31 23:45:45.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/61/844961/2/check/DellEMC_PowerStore_iSCSI/90f9878/DellEMC_PowerStore_iSCSI/677 : SUCCESS in 1h 55m 51s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/61/844961/2/check/DellEMC_PowerStore_FC/da55b58/DellEMC_PowerStore_FC/832 : SUCCESS in 1h 20m 42s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/61/844961/2/check/DellEMC_PowerStore_NFS/4b4be51/DellEMC_PowerStore_NFS/368 : ABORTED in 3h 00m 01s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/61/844961/2/check/DellEMC_PowerStore_NVMeTCP/a3160e8/DellEMC_PowerStore_NVMeTCP/242 : SUCCESS in 1h 54m 12s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"3b9d395a5d22e7112b1437c09acebdc73263c885","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-09-01 02:54:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/61/844961/2/check/dsvm-tempest-cisco-zonemanager-job/fcf639e/ : FAILURE in 1h 06m 02s","accounts_in_message":[],"_revision_number":2},{"id":"379582926f98795874dabff2366dbba09a0fdc9d","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-09-01 14:48:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/844961-2/HPMSA-iSCSI/ : SUCCESS in 54:51.95\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/844961-2/HPMSA-FC/ : SUCCESS in 54:38.08","accounts_in_message":[],"_revision_number":2},{"id":"21171d8cec3ca15b7f5a8630c6a8fef3b0756b19","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-09-01 23:51:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/844961-2/PVME-iSCSI/ : SUCCESS in 1:09:06\n- PVME-FC https://os-logs.seagate.com/dell-ci/844961-2/PVME-FC/ : SUCCESS in 1:07:40","accounts_in_message":[],"_revision_number":2},{"id":"9d0d8f6e9ea1a4c52cc3c640d609f8431e7d16ac","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-09-01 23:51:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/844961-2/Lenovo-iSCSI/ : SUCCESS in 1:07:51\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/844961-2/Lenovo-FC/ : SUCCESS in 1:08:13","accounts_in_message":[],"_revision_number":2},{"id":"facc1093115313c4bbf68148e8abb70178ae93e4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-02 03:12:42.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/61/844961/2/check/DellEMC_Unity_FC/1ec24b2/DellEMC_Unity_FC/1640 : SUCCESS in 1h 19m 46s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/61/844961/2/check/DellEMC_Unity_iSCSI/04b2e00/DellEMC_Unity_iSCSI/1530 : FAILURE in 1s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"c21c834107c347ec8bc5527daa9dc04286da22c6","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-09-02 08:17:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/844961-2/Seagate-iSCSI/ : SUCCESS in 1:16:48\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/844961-2/Seagate-FC/ : SUCCESS in 1:19:17","accounts_in_message":[],"_revision_number":2},{"id":"fc7b4426b273604e6e57eeffb3d09430a09ed9f4","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-09-07 06:55:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/61/844961/2/20220907091231/NEC-ISCSI : SUCCESS in 3h 35m 49s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/61/844961/2/20220907091231/NEC-FC : SUCCESS in 6h 42m 28s","accounts_in_message":[],"_revision_number":2},{"id":"39c9d424a8e4c62f53d1216eca93a89b813136f7","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-09-09 10:48:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/61/844961/2/ : SUCCESS in 1h 27m 25s","accounts_in_message":[],"_revision_number":2},{"id":"ff3f874e9d0ea23aa00e3320ff824741cb6cc780","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-09-13 09:34:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4171de8d58bf996b0051433a0b9357434846ce05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 09:35:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1c3478a24fb31dfcc645a791980be20504095141","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-09-13 09:59:55.000000000","message":"Patch Set 2: -Code-Review -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dbed81a9093c99a2c8cd7a768fd7f77eb782b6b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 10:00:08.000000000","message":"Patch Set 2: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":2},{"id":"fc315ed880766e2773d2c5a400ef5ff73f18ed80","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-13 11:49:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"efb10e21a039591fa2da902c9d8e9373688e0cfb","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-09-13 12:10:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6f4cf2ab13bfa42350297b9a91bc9e811719046","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-09-13 13:40:16.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/d846de4ddd1d41ba93f641c31f319f45 : FAILURE in 1h 29m 00s","accounts_in_message":[],"_revision_number":2},{"id":"b3f8e3262ba4fb2d9256a5f9671ffc28363fe09f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 14:21:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7427fdd9f024d57ae6d500b98d858b2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de095ccfead4dadb5cb9bc62738368f : SUCCESS in 9m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71fe9b69208f4a67a161af68879516e5 : SUCCESS in 12m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95031524b0b94d6b91d0ddbd93fb3a5f : SUCCESS in 12m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9bb3f6fc8ed84e91b88580c8127effac : SUCCESS in 15m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03825a42f3a34fcabcbfc59c862fdbfb : SUCCESS in 12m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/4c545d08922a4fe78e71c16e2e06b570 : SUCCESS in 54m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cd44cda78fe649178278614da14b8b88 : SUCCESS in 1h 25m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec2632bc977a4241bb9108e48a8ca3c6 : SUCCESS in 1h 24m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/003510deac9b4c34b189294d9e8bf8a6 : SUCCESS in 17m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/886727ec2eb44701a7501a1f92762a7b : SUCCESS in 5m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/662abee8c82e4f28885f6e8fe9389707 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f3896ce8f5c2459291cbc4f042201c26 : SUCCESS in 7m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/88bbb54367324986ad715cc830863b14 : SUCCESS in 7m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2310563d3e30427cb43ed8f5a038772e : SUCCESS in 48m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/32151f9e2fbd473196f8f949d769b68b : FAILURE in 7m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e97405f52f8146f6a9550eec692becc7 : SUCCESS in 1h 22m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/68da0b57c0a74bffbfd0ad6f9ba72f78 : FAILURE in 1h 17m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b317e4158a65433d9a541ff4c630dc5c : SUCCESS in 1h 48m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a01fd4ed96554ec192dd360b68934b2e : SUCCESS in 1h 09m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b722ee8816e848b79a4db2540642771c : SUCCESS in 1h 25m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/913a8ee5f327484588394bd4124bfb21 : SUCCESS in 34m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c6a611aa5b31455fac0af08574c98a68 : SUCCESS in 1h 40m 08s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cd1b7dde661f45a783533caf681f6eea : SUCCESS in 1h 59m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0743175ff1ee49a98248a0d8d4ca1471 : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":2},{"id":"2608dced045326e7dc8daf45b1af53ddf31fc8fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 14:22:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"48d5cda8d85f9108e569400c4bff4d2074bcdf92","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-09-13 14:29:27.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_09_2022_14_21_31\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"b7d905c26e99fabbb9f7f21ccd5f1e2c6337912b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 15:43:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f869b88c4ad44fa898f330860859761\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c14703c27ee14ad9a52ee5de5f361056 : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0e3b154cc1a4759b9a8129cfb0f0be4 : SUCCESS in 10m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bcf7621b25bc47539cf1186cb896d325 : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeb26a9be495406cbdf42a94103ed633 : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/6866ce8dc7184a409d17a12ec3bcb57a : SUCCESS in 56m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3a2be62b3cf44fabad4e381e3d8519d4 : SUCCESS in 1h 15m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1b944d3678d4f2ba3457be2ef7c086e : SUCCESS in 1h 19m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bd4798271aa44a5ba6c7e3686ca47f73 : SUCCESS in 49m 45s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eee2ea5f03ab4caaa6a545af0df9fa97 : SUCCESS in 1h 14m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/649cc5dc7af7442a95cf4fdc3f70046c : SUCCESS in 54m 48s","accounts_in_message":[],"_revision_number":2},{"id":"ea269c48a71cd51a6ff7c71581d4867a9c80db22","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 15:43:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"26b06236b821742406b9d70399d925b3b7ed76ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 15:44:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f697ca367a0f4501ad45fa90320405aa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ba3c3fcf1764e0e995d8963eb1c686c : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2},{"id":"86f5032c8f7c7ad41577510a653707efdfec5647","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-13 15:51:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/61/844961/2/upstream-check/cinder-tempest-plugin-ontap-fcp/75a52a2/ : SUCCESS in 1h 13m 38s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/61/844961/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/9ded129/ : SUCCESS in 46m 56s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/61/844961/2/upstream-check/cinder-tempest-plugin-ontap-nfs/80e7202/ : SUCCESS in 1h 19m 44s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/61/844961/2/upstream-check/cinder-tempest-plugin-solidfire-iscsi/7147f49/ : SUCCESS in 1h 28m 43s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"5bf9e7dba60af4918d369195d6cdf1fe6c9f253d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-09-13 16:21:43.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/61/844961/2/check/DellEMC_PowerFlex/a784ee8/DellEMC_PowerFlex/1311 : ABORTED in 2h 00m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"3629dde142a77f4383c3c11faacd379b16ad4a5d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-09-13 16:30:10.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/61/844961/2/check/DellEMC_PowerStore_iSCSI/f6d573e/DellEMC_PowerStore_iSCSI/752 : SUCCESS in 2h 08m 25s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/61/844961/2/check/DellEMC_PowerStore_FC/b863021/DellEMC_PowerStore_FC/906 : SUCCESS in 1h 17m 06s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/61/844961/2/check/DellEMC_PowerStore_NFS/054c4f3/DellEMC_PowerStore_NFS/441 : FAILURE in 1h 39m 36s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/61/844961/2/check/DellEMC_PowerStore_NVMeTCP/1f9dfe4/DellEMC_PowerStore_NVMeTCP/316 : SUCCESS in 1h 52m 12s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"7db2b8c7614317d0bf55a6d86a3b076d48e58a5a","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-09-13 20:58:07.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/844961-2/Lenovo-iSCSI/ : SUCCESS in 1:13:04\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/844961-2/Lenovo-FC/ : SUCCESS in 1:06:37","accounts_in_message":[],"_revision_number":2},{"id":"59d7756697a1cc98325460cddba517b0625437c3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-14 01:26:58.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/61/844961/2/check/DellEMC_Unity_FC/8d566bb/DellEMC_Unity_FC/1771 : SUCCESS in 1h 18m 59s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/61/844961/2/check/DellEMC_Unity_iSCSI/1daef39/DellEMC_Unity_iSCSI/1623 : FAILURE in 1h 57m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"59b108b7b06543573d3acda2ea1f7377d77bf2a2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-14 07:45:41.000000000","message":"Patch Set 2:\n\nUNSTABLE build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/610\u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"dfc1fecc0f9eafd83c8a61d445604ada43406608","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-14 09:32:42.000000000","message":"Patch Set 2:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/616 \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"8ae5a96e5fb692ee9f8a25a0b14874cfde5c94bf","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-09-14 13:17:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"68d1ce1084a0c5c8929eb3afa6504b39fc175556","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-14 17:41:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22e9e38670b6670fccf3ae7f76edcede58e0f3c4","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-09-14 20:28:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/844961-2/HPMSA-iSCSI/ : SUCCESS in 54:23.08\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/844961-2/HPMSA-FC/ : SUCCESS in 52:40.61","accounts_in_message":[],"_revision_number":2},{"id":"dbbd1aadb0a7319e03b3d4140037af5e0029e5b6","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-09-15 17:26:00.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/844961-2/Seagate-iSCSI/ : SUCCESS in 1:08:15\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/844961-2/Seagate-FC/ : SUCCESS in 1:09:06","accounts_in_message":[],"_revision_number":2},{"id":"3211495ffba8c7725db94ce9bc2cfbbebe7981bc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-09-16 11:24:58.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/50e8720b2dec468aa9e9b19f77b819c2 : SUCCESS in 39m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0eb2d1f0a73a4c6e3363064945ac2b58f55e4057","revisions":{"d9321f263ee31ab1536e6265721e56e58bbf4805":{"kind":"REWORK","_number":1,"created":"2022-06-07 12:23:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/61/844961/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/61/844961/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/844961/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/844961/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/844961/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/61/844961/1"}}},"commit":{"parents":[{"commit":"43c1f3107ed99732cecdb7a448edee16749722c9","subject":"db: Migrate \"driver initiator data\" APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43c1f3107ed99732cecdb7a448edee16749722c9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-02-17 16:05:17.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-07 12:22:52.000000000","tz":120},"subject":"db: Remove unnecessary engine facade decorator","message":"db: Remove unnecessary engine facade decorator\n\nWe\u0027re using a context manager in one function so there\u0027s no need to wrap\nthat function in a decorator.\n\nWhile we\u0027re here, we add tests to prove that the decorator is sufficient\nelsewhere.\n\nChange-Id: I3be92fcd4ac1f86230adf7c2998e2f0100a2ec33\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d9321f263ee31ab1536e6265721e56e58bbf4805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d9321f263ee31ab1536e6265721e56e58bbf4805"}]},"branch":"refs/heads/master"},"0eb2d1f0a73a4c6e3363064945ac2b58f55e4057":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-31 08:56:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/61/844961/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/61/844961/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/844961/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/844961/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/844961/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/61/844961/2"}}},"commit":{"parents":[{"commit":"743385ac1935bdd38a5f3d77e171afeb88473c1b","subject":"Merge \"Tests: Monkey patch before logs are loaded\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/743385ac1935bdd38a5f3d77e171afeb88473c1b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-02-17 16:05:17.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-08-31 08:52:56.000000000","tz":60},"subject":"db: Remove unnecessary engine facade decorator","message":"db: Remove unnecessary engine facade decorator\n\nWe\u0027re using a context manager in one function so there\u0027s no need to wrap\nthat function in a decorator.\n\nWhile we\u0027re here, we add tests to prove that the decorator is sufficient\nelsewhere.\n\nChange-Id: I3be92fcd4ac1f86230adf7c2998e2f0100a2ec33\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0eb2d1f0a73a4c6e3363064945ac2b58f55e4057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0eb2d1f0a73a4c6e3363064945ac2b58f55e4057"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
