)]}'
{"id":"openstack%2Fcinder~846054","triplet_id":"openstack%2Fcinder~master~Idf123c4fa2b3cea1cb7b5cb78c0c610b0f29dd2f","project":"openstack/cinder","branch":"master","topic":"cinderlib-rbd-keyring","attention_set":{"13671":{"account":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"last_update":"2023-11-13 17:09:29.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-08-24 11:55:45.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-11-15 16:16:55.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}},"hashtags":[],"change_id":"Idf123c4fa2b3cea1cb7b5cb78c0c610b0f29dd2f","subject":"Fix cinderlib RBD keyring issue","status":"NEW","created":"2022-06-15 18:09:41.000000000","updated":"2023-11-15 18:07:39.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":6,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"cb38bec8f9cd0fb77917d96e27df95179daaadbb","_number":846054,"virtual_id_number":846054,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-13 19:32:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-11-15 16:16:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2023-11-15 16:16:55.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-15 18:34:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-15 18:37:40.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-06-15 18:41:20.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-15 19:03:52.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-15 19:15:58.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-06-15 19:45:16.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-06-15 20:36:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-15 21:45:00.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-15 22:17:38.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-15 22:57:56.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-06-15 23:27:21.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-06-16 02:54:06.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-08 20:38:38.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-07-13 14:04:30.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-24 14:15:48.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-25 05:11:11.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-25 12:56:11.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2022-08-26 14:15:03.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-08-27 01:43:32.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-09-29 14:01:58.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2022-10-21 15:51:02.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"state":"REVIEWER"},{"updated":"2023-11-13 22:09:32.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-11-14 07:24:58.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-11-14 11:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-11-14 22:22:38.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-11-15 00:47:39.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-11-15 16:16:55.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-11-15 18:07:39.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"}],"messages":[{"id":"33be7db32cdb4f515ec81f93a0c73e8b4411b215","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-06-15 18:09:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81a0c3a0bf07adb60add8b9b5b77226c1084cc8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 18:34:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/543cab488377493a8d4b779391f71e15 : SUCCESS in 22m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fbd2e142003346bfa062073a6c1a0133 : SUCCESS in 13m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b6717717ccc1859c07aac9c0df0cb3b325a78da","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-15 18:37:40.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1655316590701 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"81c847aeedb5ddf1b68b3b52b57d0b79fc6b279b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-06-15 18:41:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/54/846054/1/ : RETRY_LIMIT in 8s","accounts_in_message":[],"_revision_number":1},{"id":"e2552addfcbf0e44127b708e3fe9c38590555436","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-15 19:03:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/149dd5b04a024e53acbe7b4ac318f15e : SUCCESS in 38m 26s","accounts_in_message":[],"_revision_number":1},{"id":"815c1a16260f8ba96c3d2243c7edfced2349437d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-15 19:15:58.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/846054/1/iscsi/ : FAILURE in 49m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/846054/1/smb/ : SUCCESS in 1h 05m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9966c993d1d84f9838a911ef868baf64bb618a32","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-15 19:45:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k-focal https://hpe-storage.github.io/alletra-6k-ci-logs/logs/54/846054/1/check/iscsi-driver-alletra-6k-focal/285bd1f/285bd1f.html : FAILURE in 1h 34m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a715b767f15fb98772bd09c9628d342a9f81bf33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 20:36:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8da0d0969144a499ffe738e77c2e670 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a842c1541f454615a8f21e7bef06dbde : SUCCESS in 12m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6e6f15a6e5840b9965683805039067f : SUCCESS in 12m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2aa681eb984a4b85ba04288c42059b10 : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68a0aa61512e42b18d3b71b920bb70db : SUCCESS in 18m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/4874fc61ec0c440c97efcc9f56c16032 : SUCCESS in 51m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b8ca662934904d039b40b76f264a6806 : SUCCESS in 1h 39m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8fb3b2bb4b6c41be9b9cea00f46c25fa : SUCCESS in 51m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/51ba6eba2539441db38152c95ede7cdd : SUCCESS in 15m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b799167db25247e089454efe93bc82b7 : FAILURE in 7m 10s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a642222943e84d47887fa1c597098bbc : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1d9a901df1c140ea835ef91e34063c75 : SUCCESS in 6m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b95154caada548d582d6b3364b8ad720 : SUCCESS in 6m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9811bed10d6c4c3381cf6bd1518fa512 : SUCCESS in 38m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c88d471188cb46029fcd328fedf1561c : FAILURE in 6m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/20e2ae8be4f248a1ae0658a4d86cbdb3 : FAILURE in 1h 18m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d0d5070ce98c4297a360d2dfb6c9223f : FAILURE in 1h 13m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2ef58f38cef9436cb779aaf86a5b0650 : SUCCESS in 1h 30m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0560bceb0c564057bc86609091121916 : SUCCESS in 1h 12m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e8f4b0d4db634b41bbbfd1d35ba09811 : SUCCESS in 1h 24m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/da7f04887ca244d5ac2e56914b020813 : SUCCESS in 47m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5b0c7b0e3fec48098551400911d58858 : FAILURE in 1h 10m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2eefb17e672a48d49ff4b764aabcb487 : SUCCESS in 2h 22m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bac925a2d994513bf3acb1933213b9a : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c71900cd8e1e1664f5b7dfed96ea850e4cf2b8a2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-15 21:45:00.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-54-846054-1/ns5_nfs : SUCCESS in 3h 26m 24s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-54-846054-1/ns5_iscsi : SUCCESS in 3h 34m 47s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"40ab63c7c110aff5cd151b6886e50189cc7c0203","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-15 22:17:38.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/18a582eb06f246ed9ffd6d018721a04b : POST_FAILURE in 1h 51m 27s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/341e8a31ca0a4e4ea5fb1a6d2edf1239 : TIMED_OUT in 3h 06m 19s","accounts_in_message":[],"_revision_number":1},{"id":"274be84441302e81d53ff66c78e7141a7fb42865","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-15 22:46:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/846054/1/check/iscsi-driver-nimble/5ea83aa/5ea83aa.html : FAILURE in 1h 31m 33s","accounts_in_message":[],"_revision_number":1},{"id":"8775ad9b5f84b3f9521e8f993cf5d7ecffd09795","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-15 22:57:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/846054/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/592a7ca/ : TIMED_OUT in 3h 02m 56s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/846054/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/4e422b2/ : FAILURE in 1h 04m 16s","accounts_in_message":[],"_revision_number":1},{"id":"7c4fd809c7bbbb057c483268393bf903c6bd63bb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-15 23:27:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/54/846054/1/check/dsvm-tempest-cisco-zonemanager-job/e9e3bfc/ : FAILURE in 1h 26m 14s","accounts_in_message":[],"_revision_number":1},{"id":"3e843ce99166176d38e40807ee8fdbab15a9ae8a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-16 02:54:06.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"eed8fdd06de60a3cad9b6268bc767226a8959d1a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-17 08:42:16.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/846054/846054/1/check-not-vote/ext-cinder-zuul/0fb5cfe : FAILURE in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d5c679818047de860c6a3474e16c0f97457c7200","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-08 20:38:38.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/846054/1/20220708233526/NEC-V-ISCSI : SUCCESS in 2h 42m 15s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/846054/1/20220708233526/NEC-V-FC : FAILURE in 6h 03m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2af1e0c2b7535f77e08d7e2b4f66a98b278cea04","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-07-13 14:04:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/846054/1/20220713162431/NEC-FC : SUCCESS in 3h 25m 56s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/846054/1/20220713162431/NEC-ISCSI : SUCCESS in 6h 39m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1be20d892981e3ebb354a46bc7fcaa50ea5647db","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-24 09:52:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8fc3ddca7033df065b8d31b424e3c1dc4ed4ee7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 10:18:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/841c72ea43ff43b2bd792e36310277a7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ef46e8e17b64d80899d433c9b421ee0 : SUCCESS in 22m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a278e71951554dafb68f95ff21db78f1 : SUCCESS in 23m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8f12aa6c4e97c4518335cc5441dc07789a17e5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 11:55:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/33e0e7318ed14ed19d18c076faa2e83d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/870830a8de1d452d83e3ab469faa2f81 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e9c672dcd3f4c52bbd5dd28403cf2a7 : SUCCESS in 10m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c4fdae3a93f40b5ba901f4885a85f1b : SUCCESS in 10m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1d5792d7d0d454ba8f486a9c70a36b6 : SUCCESS in 13m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3722a5da3dd44108a30e735502f7d80 : SUCCESS in 12m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/c6341fae84c74cebbeab3154b9d7e884 : SUCCESS in 58m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d2c4193b66d94618af4e354c505aba87 : SUCCESS in 1h 20m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46cac27a97924b1dbb21d44d5dc1b2c2 : SUCCESS in 56m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b2a8e0a5bcfe4c08931603760215c1b0 : SUCCESS in 13m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/71444517ba734ebbafa7113f44df63f5 : FAILURE in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/282c489336984c009f2a725e18742c22 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/145b4e2aa909468d85e95d0768f27097 : SUCCESS in 6m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6b0143d6367842c9b017a0b3d42e7639 : SUCCESS in 7m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/758fa74e61624bd1b438340e6d51d7f9 : SUCCESS in 39m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2f0468dcfb2e434fb24e4c8e53b4d62d : FAILURE in 6m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/59dee12e77e1402289847fcdfd9d491e : SUCCESS in 1h 19m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b65e0d5f8a8a4c67baf63d6084642051 : FAILURE in 1h 27m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a34567f116e04fb2baa2a8a11114f0bc : SUCCESS in 2h 01m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5d344c13afe047108f430a7316455a0e : SUCCESS in 1h 05m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3c4cf0b643b04b029c92d9f7e9ad3d24 : SUCCESS in 1h 30m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/92a16f64e4344e5e8485aa091f477cc8 : SUCCESS in 38m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/abfc113d95074d868b027c56cf5c65a9 : SUCCESS in 1h 23m 13s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad99228c10cd436785d21964ae656c70 : SUCCESS in 1h 54m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f2350cb50644dfe8c2f7bb743f03809 : SUCCESS in 51m 36s","accounts_in_message":[],"_revision_number":1},{"id":"77d5111a9637a4d35fd6b3c26d5b1dfd5e9beb2e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-24 12:47:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/846054/1/iscsi/ : SUCCESS in 54m 54s\n- cinder-smb http://cloudbase-ci.com/cinder/846054/1/smb/ : SUCCESS in 52m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4ce83ca72c811012d4a1199a9ff3245da3799249","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-24 13:49:11.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/846054/1/check/tempest-dsvm-full-focal-py3/703d326/ : SUCCESS in 2h 01m 40s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/846054/1/check/devstack-plugin-ceph-dsvm-focal-py3/16fd2dc/ : SUCCESS in 2h 40m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3dc15919664eb147266c2c5e9ba88221710f8bab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-24 14:15:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8672d049a16c7d9975d53e0c4ff8fd663acf6662","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-24 14:15:48.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"382570250e58c40c3ea4042ef6aa4a2870760736","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-24 14:16:24.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"fa84409813689a610779048c9cddf9921eb7c2a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 15:17:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a2a777772874d0084e962750eeeedcd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3eb2301f6e24a24a532b70ffa9e7430 : SUCCESS in 19m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0df008ac7bfc4e529ae1453fbb9efe80 : SUCCESS in 22m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"027486165ccef15f0c2210791370e2f6402ae1ba","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-24 15:57:10.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/846054/2/iscsi/ : FAILURE in 53m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/846054/2/smb/ : SUCCESS in 55m 23s","accounts_in_message":[],"_revision_number":2},{"id":"b4d4b538b3243849fe370c422966ce3bd8b483cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 16:18:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37b407d896684132960881aae7e130c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2474541a2e9c48ac80bb3046df615777 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5446e2d124dc4b0cb619d575a47fec31 : FAILURE in 24m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b719c90aa4941a38f447641322bcbd1 : SUCCESS in 13m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58d9d9eacff24079beedcd62f95753bd : SUCCESS in 13m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f326cb74d354df9ba575eb79b35c6a2 : SUCCESS in 12m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/f767b28b7766487ab2b47a80a789f153 : SUCCESS in 59m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/61059e81983f420f8660b4c5b4796587 : SUCCESS in 1h 16m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9c1460b2eb441a0bb285bfa7f44fa39 : SUCCESS in 52m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d07ad21984d940de969b862ae6ea6f31 : SUCCESS in 13m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/62c10b2cb0804c05b9c16ff35c65c46a : SUCCESS in 7m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/264e0d1c02e041369b260737ad1b3a41 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a08a1f05b2334b5cad6276f5fd61369d : SUCCESS in 6m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f71644d34a8d47189f0d52bd0e181cf5 : SUCCESS in 8m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/36d3effe34984077b962b618367a5bc7 : SUCCESS in 46m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4193d82d388848dda42af1a74e72fb12 : FAILURE in 6m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4f54872c56504f5bae947ed3747f0058 : SUCCESS in 1h 22m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/eb3fd84654aa4702a98316d03a5f5519 : FAILURE in 1h 44m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4bd161e443c74680889efa4880e1760f : SUCCESS in 1h 43m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d87ff6e9c0844ed18403816589b8111d : SUCCESS in 1h 10m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/47281f7fe18d4fb5a265644cc6425dce : SUCCESS in 1h 29m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3e55cdc9e1a0407ebc55de555f7118f8 : SUCCESS in 44m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/04cfa865abac42d987a04c91e79f83ff : SUCCESS in 1h 30m 07s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5cae277798754905bb0378553c07b315 : SUCCESS in 2h 01m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0cbfac088da74c76acc6c40ee8f19039 : SUCCESS in 53m 14s","accounts_in_message":[],"_revision_number":2},{"id":"8c7f50232f8ee12ee0e1522cdfe95940d0ba1d5b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-24 17:05:25.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661350510580 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"ec1642aed266baa0ea0d72100635709342745e60","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-24 17:49:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/54/846054/2/check/dsvm-tempest-cisco-zonemanager-job/52c42c7/ : POST_FAILURE in 12m 32s","accounts_in_message":[],"_revision_number":2},{"id":"9800f954950a0a7ec2becf7c8bae8f2380a78220","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-24 18:15:52.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/54/846054/2/check/iscsi-driver-alletra-6k/de7050e/de7050e.html : FAILURE in 1h 05m 42s","accounts_in_message":[],"_revision_number":2},{"id":"dc4835d83431c0c55eed8b4d6d8857df4ce7a9f3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-24 19:29:56.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-54-846054-2/ns5_nfs : SUCCESS in 3h 28m 10s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-54-846054-2/ns5_iscsi : SUCCESS in 3h 39m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"9f8953895a4c1431dd9e61851e822de9f98a0b5a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-24 20:51:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/846054/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/25bc0da/ : SUCCESS in 1h 41m 55s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/846054/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6cfbb22/ : SUCCESS in 1h 41m 56s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/54/846054/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/b0fa6f7/ : SUCCESS in 1h 45m 15s","accounts_in_message":[],"_revision_number":2},{"id":"82cb6a8dbe1bc96ca2cbd531974bd9480861df6d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-24 22:31:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/846054 : SUCCESS in 1h 38m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2c82db4ba0aab170b20f3a77dd23e915aab2d189","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-25 05:11:11.000000000","message":"Patch Set 2:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/54/846054/2/check/fujitsu-eternus-dx-iscsi/17da503 : SUCCESS in 1h 37m 00s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/54/846054/2/check/fujitsu-eternus-dx-fc/63325a0 : SUCCESS in 1h 36m 36s","accounts_in_message":[],"_revision_number":2},{"id":"62ba3533b821fae3975b997abbf3fc2727e83cee","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-25 12:56:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"51af729479f722113934d96d2d92eef93c480db6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-26 14:15:03.000000000","message":"Patch Set 2:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/488","accounts_in_message":[],"_revision_number":2},{"id":"9013f51f504a7135f657fe8146fee871b9006c2d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-27 01:43:32.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/54/846054/2/check/DellEMC_Unity_FC/ccf6329/DellEMC_Unity_FC/1565 : FAILURE in 14m 14s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/54/846054/2/check/DellEMC_Unity_iSCSI/a53c7ad/DellEMC_Unity_iSCSI/1440 : SUCCESS in 1h 49m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"8ce57a4b194eaaae7e4177db15063be93081a65c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:36:08.000000000","message":"Patch Set 2:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":2},{"id":"0b101b7c42f0458ec1f2784712e539f95fd22d68","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-30 09:24:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/846054/2/20220830145223/NEC-FC : SUCCESS in 3h 27m 19s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/846054/2/20220830145223/NEC-ISCSI : SUCCESS in 3h 32m 19s","accounts_in_message":[],"_revision_number":2},{"id":"35d37043c2b394085638d79d30b1ef1717072963","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-09-05 22:43:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/846054/2/20220906013408/NEC-V-ISCSI : SUCCESS in 2h 52m 52s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/846054/2/20220906013408/NEC-V-FC : SUCCESS in 6h 09m 28s","accounts_in_message":[],"_revision_number":2},{"id":"8e8d7772eb21b801b2004935cc48ad111ff2586f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-09-14 23:54:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/adc1aba4034140aea4f865a616a0d4cc : SUCCESS in 39m 50s","accounts_in_message":[],"_revision_number":2},{"id":"10fd7d75e6ab00b3a57832bf4224f4ca8b7672b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-09-29 12:53:29.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"84c8755e1fb9bbc085977482dc833f3058288aab","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-09-29 13:07:58.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/846054/3/iscsi/ : FAILURE in 13m 57s\n- cinder-smb http://cloudbase-ci.com/cinder/846054/3/smb/ : FAILURE in 13m 54s","accounts_in_message":[],"_revision_number":3},{"id":"d8bff773f58335ae1950a97e2608891e1b4c3b48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 13:25:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5065d08921664fd6a054660c0e545ea1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6c8062cfcfb4403ea8f74218582cae93 : SUCCESS in 28m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/358ff5d645b247f19794eaf912bb06c4 : SUCCESS in 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d6882d5582fdf44134a4b2f38e5af6df6a666815","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-09-29 14:01:58.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/2a7f102e60b44e40928bb23f48648e2e : FAILURE in 1h 06m 08s","accounts_in_message":[],"_revision_number":3},{"id":"703b93782c177393eb7d34d38c9a102b8929d876","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-29 14:05:03.000000000","message":"Patch Set 3:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/710","accounts_in_message":[],"_revision_number":3},{"id":"6c3f4d9421a2e9b3a5f99299c9c4e171c9c54b91","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-09-29 14:32:42.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/54/846054/3/check/iscsi-driver-alletra-6k/14beefb/14beefb.html : FAILURE in 1h 03m 07s","accounts_in_message":[],"_revision_number":3},{"id":"41a3f51eb63682e1949a8780df7bd72329a79e0d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-29 14:40:16.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/54/846054/3/check/DellEMC_Unity_FC/52cdd39/DellEMC_Unity_FC/1877 : SUCCESS in 1h 07m 29s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/54/846054/3/check/DellEMC_Unity_iSCSI/34f7f1b/DellEMC_Unity_iSCSI/1724 : SUCCESS in 1h 45m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"1f98bcfc1a507721ce76039a40df3aa4c8537703","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-29 14:58:12.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/846054 : FAILURE in 30m 44s","accounts_in_message":[],"_revision_number":3},{"id":"f83b789620c31bea6df17cc3d908de73fcb696a9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-09-29 15:07:09.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/54/846054/3/check/dsvm-tempest-cisco-zonemanager-job/7315565/ : FAILURE in 1h 11m 36s","accounts_in_message":[],"_revision_number":3},{"id":"0d9eb29d0dd7e5b6d5cbafd903c9928325d1f142","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:12:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/434431f9ba4b4df6b111e0ba1290ce9b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/464cf3a7db4940ca97c292bc7e0cdb54 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9c5b3157f2a4b41994f4dc580cc0dc1 : SUCCESS in 10m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a754b3afeb14e8d933ae22727d12765 : SUCCESS in 9m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c03deb6a79844a88b5d493e86be4cd41 : SUCCESS in 10m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/ee14ad950bad4cbdbe9129dbbd07fa02 : SUCCESS in 58m 54s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bc27780796ff4ad68a740aab31a9c8a7 : SUCCESS in 47m 02s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e2327cbc8928440d93b8d96800c78e2a : SUCCESS in 1h 06m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e11be66b3be04fdab92845ab0b92e6aa : SUCCESS in 43m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/29e88d54dd054512a1576d09e94ec278 : SUCCESS in 17m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/07a2db54b7594502b4aefa011e797f08 : SUCCESS in 5m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/43af0da2647f46d3951778a0072e291e : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e55e52f9c33b448eba0becbacc04cc36 : SUCCESS in 7m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/34906f8043dd4e0ba3ec955f3a0f78f2 : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3b847504b03e412b906ddfabf712f23d : SUCCESS in 46m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0bb12c935f164dd38f7f31b199857325 : FAILURE in 8m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1f7e54cca4924f4093c1a34b85a17201 : TIMED_OUT in 2h 06m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/87c75e00656246b88b1176f914826cfa : TIMED_OUT in 3h 06m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6d1eb45922c94e79ae7def29fc2120d1 : SUCCESS in 1h 38m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7c1ecb8520f74be58655bc5693ecddfb : SUCCESS in 55m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4af465efb7a848afae7e8841916752fb : SUCCESS in 1h 26m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/acd20e7e590c42b6ac53b1ced6e52cfc : SUCCESS in 35m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/71a4f7cf962a4ffcae056dc5fce272d6 : SUCCESS in 1h 30m 25s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eb1336b28c5a45c2b271f2c3300b0a7d : SUCCESS in 2h 08m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b00314a17ee14cfaa04981290c46ed83 : SUCCESS in 46m 20s","accounts_in_message":[],"_revision_number":3},{"id":"0f57e6741f0fb56b14bd907c990522cba4f12b51","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-29 16:39:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/846054 : SUCCESS in 1h 28m 47s","accounts_in_message":[],"_revision_number":3},{"id":"6dda39a15669a6fd6155ac0a816e0674c7b622d7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-09-29 17:31:46.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1664456019150 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"c7d161ab46ac53effff4cf2952169a77079c8ebb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-29 18:06:04.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-54-846054-3/ns5_nfs : FAILURE in 3h 25m 34s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-54-846054-3/ns5_iscsi : SUCCESS in 3h 28m 20s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"a1b3ed05289825e829731d016ee26c93c9ef8a59","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-29 20:51:46.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/846054/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d176fe0/ : SUCCESS in 1h 38m 41s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/846054/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/059f4bc/ : SUCCESS in 1h 44m 57s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/54/846054/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/c0dab0d/ : SUCCESS in 1h 38m 47s","accounts_in_message":[],"_revision_number":3},{"id":"0a95d6cf244d85712329164084e8761decc1d429","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-09-30 23:10:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/846054/3/20221001043641/NEC-FC : SUCCESS in 3h 32m 19s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/846054/3/20221001043641/NEC-ISCSI : SUCCESS in 3h 33m 26s","accounts_in_message":[],"_revision_number":3},{"id":"b304d2ffc22b9273a472d5754e0f69c9fda25eb2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-10-16 23:17:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/846054/3/20221017022606/NEC-V-ISCSI : SUCCESS in 2h 19m 58s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/846054/3/20221017022606/NEC-V-FC : SUCCESS in 5h 51m 47s","accounts_in_message":[],"_revision_number":3},{"id":"2a5bb098bcd66686835ecb62321737726ff46c04","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-10-21 15:51:02.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e6922d9c064ed518248d4cbb5684e4954b5d19ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-11-13 17:09:29.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"74c528e62b13c971c91ec56f07abd56715632891","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 17:47:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaa36d3c1e8e4523b9103244365f9552\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5bfed0dadaee423b8412750d004bdde2 : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3ec15f2f1a9e42e39f455003612717f1 : FAILURE in 27m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/53cf71af4a68496084fc5ae9c7e52003 : FAILURE in 27m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9a5852e487f27b38fd4c12b6ae7d498bafecad92","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-13 17:52:20.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8acd46cb77394d76a8e36b16160c2afa : RETRY_LIMIT in 5m 12s","accounts_in_message":[],"_revision_number":4},{"id":"42e6a3922587060f8a5da743022b425dc22fc4fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 19:32:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5816b9dfe241484882d9cd6db4d734a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7a0024d1d584d54a2dc0dbcc467193a : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3cad8981aff4736a1abd18c8c3c77cd : SUCCESS in 10m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f30e6d433e914c29a02b59fc73c5d738 : SUCCESS in 11m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cfd46460fad47979ccfd2a9a327e31d : SUCCESS in 12m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/d091839a28304a44a4ff00d1c484939b : SUCCESS in 59m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6e86596745bf436487bbc301b9b5c57b : SUCCESS in 1h 01m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/985476b2f2cd4adaa7909510df56696c : SUCCESS in 1h 37m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03bcda9b6cd94e3ba466e3aada262b53 : SUCCESS in 1h 05m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/954c073802d24a51923a3dc97ae3f4f6 : SUCCESS in 17m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c18ffddc73394bc080b771cce6a9d28a : SUCCESS in 5m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/203d8a17465049d0b9382212f9bfba48 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/13dfcdbec4dd48979a8558c9551c82ca : SUCCESS in 6m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5f094cc295a041e3b1779e7a0bb95446 : SUCCESS in 5m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0c3408cd6f034f4b96947f43da154022 : SUCCESS in 51m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f4ba579976c844bab10d39424976a7a8 : SUCCESS in 7m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1b7bc42f3fd6488b88992bfada60ac6a : SUCCESS in 1h 52m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4bde217fb73d4242a8b2c833474dac87 : SUCCESS in 2h 13m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8814522d953d4b2bb54806f0eb933056 : SUCCESS in 1h 49m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/91101798e18f4346931a440d40fbe9f2 : SUCCESS in 2h 05m 00s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6cd0181979a64500b0f651e7239252cc : SUCCESS in 1h 25m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/871bee20c1d94c8fa0644ede7bbf350d : SUCCESS in 1h 24m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cc315aaea7a84d4393c5d5e80e92f088 : SUCCESS in 50m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f813b1b72c0e47a7956af570179c2272 : SUCCESS in 1h 25m 42s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/419fc4c565174f2b90c688565a751b39 : SUCCESS in 1h 34m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/79910211335141acbeb2a20d939dcede : SUCCESS in 1h 28m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7919c90a4034bc99ff449a6aa56d183 : SUCCESS in 58m 06s","accounts_in_message":[],"_revision_number":4},{"id":"8039d4203520f891ad2d8727662888a8c0b0e57c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-13 19:40:47.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/434a2c93ade040b7b53be54c1f18ac4c : RETRY_LIMIT in 43s\n- pure-devstack-cinder-tempest-fc-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/0ee6eb7f9221484aa8b44547d6372356 : RETRY_LIMIT in 43s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/28d5ef61159542b7b088715d45420868 : RETRY_LIMIT in 42s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/ff203b873d2d4945b0bf298ccd7f4625 : RETRY_LIMIT in 42s","accounts_in_message":[],"_revision_number":4},{"id":"ff4649223e319f7d8c53a1fc649d5452d564780f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-11-13 22:09:32.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/846054/4/upstream-check/cinder-tempest-plugin-ontap-fcp/295a303/ : SUCCESS in 2h 36m 14s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/846054/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/3a5270b/ : RETRY_LIMIT in 8m 53s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/846054/4/upstream-check/cinder-tempest-plugin-ontap-nfs/3c2d86a/ : RETRY_LIMIT in 9m 18s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/846054/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/b033269/ : RETRY_LIMIT in 8m 16s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"8b35e4992dac903e6cb808789188352befd517f2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-11-14 00:24:19.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/f13dae73d4b9490fa34dd53f711da8e3\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/8923e4e51510463189c6df51208f8cf6 : SUCCESS in 1h 20m 55s","accounts_in_message":[],"_revision_number":4},{"id":"bd0cc0b454853456e7b406705a5a12e7acdc3735","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-14 04:21:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/66/.","accounts_in_message":[],"_revision_number":4},{"id":"67733979582116cda5848c91c80133246554912f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-14 07:24:58.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-846054-4/FC/ : FAILURE in 8h 12m 48s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-846054-4/iSCSI/ : FAILURE in 8h 30m 8s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"1840870caa65ed3b1733a488659204ac92cb6ca7","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-11-14 11:00:08.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/54/846054/4/fc : SUCCESS in 0h 48m 32s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/54/846054/4/iscsi : SUCCESS in 0h 57m 10s","accounts_in_message":[],"_revision_number":4},{"id":"499e8cec53638aeb648049c609a4c8a41d7dac62","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-11-14 17:11:39.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/54/846054/4/ : SUCCESS in 1h 34m 43s","accounts_in_message":[],"_revision_number":4},{"id":"e241cc690f7b23b30498abd147eb6e10d0036aa9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-14 19:21:55.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/86/.","accounts_in_message":[],"_revision_number":4},{"id":"5ced17869c3e5844d5c3860bce82cdb9b924b20d","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-11-14 22:22:38.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/846054-4/PVME-iSCSI/ : SUCCESS in 2:31:37\n- PVME-FC https://os-logs.seagate.com/dell-ci/846054-4/PVME-FC/ : SUCCESS in 1:52:50","accounts_in_message":[],"_revision_number":4},{"id":"80d5962db12f4243876f4f8bbd9d01f5e5df31fb","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-11-15 00:47:39.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/846054-4/HPMSA-iSCSI/ : SUCCESS in 1:57:19\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/846054-4/HPMSA-FC/ : SUCCESS in 1:34:32","accounts_in_message":[],"_revision_number":4},{"id":"9337dc0f4c5b1d77cf4b925595c320b6c469bb95","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-11-15 03:35:13.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/846054/4/20231115091400/NEC-V-ISCSI : SUCCESS in 3h 09m 08s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/846054/4/20231115091400/NEC-V-FC : SUCCESS in 3h 21m 06s","accounts_in_message":[],"_revision_number":4},{"id":"38759984310b5d772dc8d4f0bb5c8aea3e002c3a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-15 13:11:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/97/.","accounts_in_message":[],"_revision_number":4},{"id":"3fc8145b018740552ddad2c4fc5236c69462fd80","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-11-15 15:00:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/846054/4/20231115195134/NEC-ISCSI : SUCCESS in 3h 49m 41s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/846054/4/20231115195134/NEC-FC : SUCCESS in 4h 09m 12s","accounts_in_message":[],"_revision_number":4},{"id":"068cfe49f88fb82ad87166fabfda734e7a7de5ee","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-11-15 16:16:55.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cb38bec8f9cd0fb77917d96e27df95179daaadbb","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-11-15 18:07:39.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"52262f9df29029a4bcec5fcbb8f0f539901dad1f","revisions":{"2ce9d127665a1dbb6fa99fd6ef885ac201ef8025":{"kind":"REWORK","_number":1,"created":"2022-06-15 18:09:41.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/54/846054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/846054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/846054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/846054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/846054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/846054/1"}}},"commit":{"parents":[{"commit":"9a8583b2ef59f6a392b8b1a1efef8d3d270cf9ea","subject":"Merge \"PowerMax Driver - Improve error handling around deletes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9a8583b2ef59f6a392b8b1a1efef8d3d270cf9ea"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-06-15 18:05:31.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-06-15 18:05:31.000000000","tz":120},"subject":"Fix cinderlib RBD keyring issue","message":"Fix cinderlib RBD keyring issue\n\nOn Change-Id Iae63aee973932b2eef26d832a7f413a04bad0df1 we added back the\nsupport for the rbd_keyring_conf option to Cinder, but the\nfunctionality is not complete, because it only fully works when the\nconfiguration file is located under /etc/ceph.\n\nThis patch fixes support for non /etc/ceph locations.\n\nNo release note has been added since this is a functionality that is\nonly possible through cinderlib (adding the config option in cinder.conf\ndoes nothing) and we don\u0027t want to create unnecessary noise.\n\nChange-Id: Idf123c4fa2b3cea1cb7b5cb78c0c610b0f29dd2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ce9d127665a1dbb6fa99fd6ef885ac201ef8025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ce9d127665a1dbb6fa99fd6ef885ac201ef8025"}]},"branch":"refs/heads/master"},"020b025621d034ba54c24e9d3f7ee099ad10d70f":{"kind":"REWORK","_number":2,"created":"2022-08-24 14:15:01.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/54/846054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/846054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/846054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/846054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/846054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/846054/2"}}},"commit":{"parents":[{"commit":"9a8583b2ef59f6a392b8b1a1efef8d3d270cf9ea","subject":"Merge \"PowerMax Driver - Improve error handling around deletes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9a8583b2ef59f6a392b8b1a1efef8d3d270cf9ea"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-06-15 18:05:31.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-24 14:14:45.000000000","tz":120},"subject":"Fix cinderlib RBD keyring issue","message":"Fix cinderlib RBD keyring issue\n\nOn Change-Id Iae63aee973932b2eef26d832a7f413a04bad0df1 we added back the\nsupport for the rbd_keyring_conf option to Cinder, but the\nfunctionality is not complete, because it only fully works when the\nconfiguration file is located under /etc/ceph.\n\nThis patch fixes support for non /etc/ceph locations.\n\nNo release note has been added since this is a functionality that is\nonly possible through cinderlib (adding the config option in cinder.conf\ndoes nothing) and we don\u0027t want to create unnecessary noise.\n\nChange-Id: Idf123c4fa2b3cea1cb7b5cb78c0c610b0f29dd2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/020b025621d034ba54c24e9d3f7ee099ad10d70f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/020b025621d034ba54c24e9d3f7ee099ad10d70f"}]},"branch":"refs/heads/master"},"93212ac63dbeeef8b88f87de3eb5328d3dd6e711":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-09-29 12:53:29.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/54/846054/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/846054/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/846054/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/846054/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/846054/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/846054/3"}}},"commit":{"parents":[{"commit":"c94962b98e599a35e5c00bdbeb699edee19bf33e","subject":"Merge \"setup.cfg: [extras] for pip-installable driver requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c94962b98e599a35e5c00bdbeb699edee19bf33e"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-06-15 18:05:31.000000000","tz":120},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-09-29 12:53:29.000000000","tz":0},"subject":"Fix cinderlib RBD keyring issue","message":"Fix cinderlib RBD keyring issue\n\nOn Change-Id Iae63aee973932b2eef26d832a7f413a04bad0df1 we added back the\nsupport for the rbd_keyring_conf option to Cinder, but the\nfunctionality is not complete, because it only fully works when the\nconfiguration file is located under /etc/ceph.\n\nThis patch fixes support for non /etc/ceph locations.\n\nNo release note has been added since this is a functionality that is\nonly possible through cinderlib (adding the config option in cinder.conf\ndoes nothing) and we don\u0027t want to create unnecessary noise.\n\nChange-Id: Idf123c4fa2b3cea1cb7b5cb78c0c610b0f29dd2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/93212ac63dbeeef8b88f87de3eb5328d3dd6e711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/93212ac63dbeeef8b88f87de3eb5328d3dd6e711"}]},"branch":"refs/heads/master","description":"Rebase"},"52262f9df29029a4bcec5fcbb8f0f539901dad1f":{"kind":"REWORK","_number":4,"created":"2023-11-13 17:09:29.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/54/846054/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/846054/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/846054/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/846054/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/846054/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/846054/4"}}},"commit":{"parents":[{"commit":"4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6","subject":"Merge \"[Pure Storage] Unit tests and fixed variable name for Replication-Enabled Consistency Groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ee1bdaf648064adb6ad9f0e4fda6adc6ad1cbb6"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-06-15 18:05:31.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2023-11-13 17:09:20.000000000","tz":60},"subject":"Fix cinderlib RBD keyring issue","message":"Fix cinderlib RBD keyring issue\n\nOn Change-Id Iae63aee973932b2eef26d832a7f413a04bad0df1 we added back the\nsupport for the rbd_keyring_conf option to Cinder, but the\nfunctionality is not complete, because it only fully works when the\nconfiguration file is located under /etc/ceph.\n\nThis patch fixes support for non /etc/ceph locations.\n\nNo release note has been added since this is a functionality that is\nonly possible through cinderlib (adding the config option in cinder.conf\ndoes nothing) and we don\u0027t want to create unnecessary noise.\n\nChange-Id: Idf123c4fa2b3cea1cb7b5cb78c0c610b0f29dd2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/52262f9df29029a4bcec5fcbb8f0f539901dad1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/52262f9df29029a4bcec5fcbb8f0f539901dad1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
