)]}'
{"id":"openstack%2Fcinder~846173","triplet_id":"openstack%2Fcinder~master~Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4","project":"openstack/cinder","branch":"master","topic":"remove-legacyfacade","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-04 18:20:33.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-24 13:05:19.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2026-06-04 18:20:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4","subject":"db: Use cinder.db.api directly","status":"MERGED","created":"2022-06-16 13:54:36.000000000","updated":"2026-06-19 20:03:52.000000000","submitted":"2026-06-04 18:20:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846173-remove-legacyfacade","meta_rev_id":"c323ea7be25631c8d66702189296d3ba300f53be","_number":846173,"virtual_id_number":846173,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-04 18:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-05-14 04:10:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-06-04 15:28:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-06-04 15:28:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2026-05-14 04:10:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-16 15:28:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-16 17:01:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-16 19:24:50.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-16 20:02:20.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-06-16 21:31:23.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-06-16 23:18:34.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-17 00:48:09.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-17 00:52:20.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-06-17 01:05:46.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-17 05:19:40.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-06-17 08:01:25.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-18 12:26:05.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-06-20 05:26:53.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-23 23:34:36.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-06-23 23:50:29.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-06-24 00:56:48.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-06-24 00:59:17.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-24 01:24:53.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-06-24 03:44:46.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-06-24 13:44:28.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-06-24 17:15:45.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-06-24 18:35:23.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-06-26 16:45:36.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-13 12:33:29.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-07-16 09:30:18.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-18 23:37:07.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-07-28 16:44:25.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-07-28 19:15:49.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-08-31 15:10:36.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-08 19:46:51.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2023-02-22 17:25:40.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-03-31 03:11:45.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-11-22 06:09:23.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-10-24 17:40:02.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-12 17:53:08.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-05-12 18:48:37.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-12 20:19:51.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-05-12 20:25:45.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-05-14 04:10:28.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-06-04 15:28:01.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-06-04 15:28:19.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"},{"updated":"2026-06-04 18:20:44.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"eba0584b1c74194889fba1eaa9e2d0c2591c3fa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-06-16 13:54:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93cd6b8223ee03a76b7bef1950d92c0d7ef0425a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 15:28:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3f4cbfe8cfb4187854106debf0b5326 : SUCCESS in 20m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9fa4f89d983842449db3056bd9151e8a : SUCCESS in 20m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db9c213066a037074e9eb0a190ad2bbf66598573","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 17:01:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11bd8e8edcd641789283fb569f39686a : FAILURE in 11m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9038b968a1242edad2e603f37e8c06a : SUCCESS in 16m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95a7cf7bbb1b4cb58d841bd91846000e : SUCCESS in 12m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a768235a3f6048ac96f8a6f722ac8f33 : SUCCESS in 12m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da184ee1067342b5a0972b39f9c5abdc : SUCCESS in 12m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/8e14845048bc48ed9a2e60c434a91f46 : SUCCESS in 1h 05m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5e012a77976b495682e4b07818869d90 : SUCCESS in 1h 26m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5410a8c1c48f4a49b4412c7b495e5942 : SUCCESS in 55m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/39dbd659818142fea0e2179dc1f3317d : SUCCESS in 19m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8ab6f64798674669b446fac0c1f02697 : SUCCESS in 6m 50s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f69c98f199cd40989067f315a859bd63 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f972a36d233740c28cb8fdbb818f6fb1 : SUCCESS in 6m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/448c24abf5d148a38aff0ffd0087f001 : SUCCESS in 8m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7117c21c41d4451d90057706679e5b56 : SUCCESS in 38m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aab4f02d971b42cf8bd248e8def9f39e : FAILURE in 7m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d55a94b10f674faca927182907ee2e45 : SUCCESS in 1h 02m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/88070adedd4e4d1fac2b8060f6d573d7 : FAILURE in 25m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2525af859b5b47fb8f7ef01daac91c58 : SUCCESS in 1h 30m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b80dd667f7354d21be775ffb58483e89 : SUCCESS in 1h 06m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a66e7f13c468447892396cfb9cbdfbef : SUCCESS in 1h 14m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1e6159289827482e946d59b03f22ca34 : SUCCESS in 45m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8c79644364a043b0b091351f10f0245d : FAILURE in 1h 18m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e333aaf1d8b84261b83abb3602b62948 : SUCCESS in 2h 35m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9c428098cbb4d7c97c4c862ad19f139 : SUCCESS in 58m 01s\n\nWarning:\n  Comments left for invalid file cinder/db/sqlalchemy/api.py","accounts_in_message":[],"_revision_number":1},{"id":"a3cbcc0eeee49f4174cc6e95c9f6627ca6f59939","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-16 19:24:50.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/1ca5c28401524c81a50037b36b734b5f : POST_FAILURE in 2h 12m 52s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"9185bf14ee525ec8679d7791a216f93abe4e93ed","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-16 20:02:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/846173/1/iscsi/ : SUCCESS in 1h 36m 05s\n- cinder-smb http://cloudbase-ci.com/cinder/846173/1/smb/ : SUCCESS in 1h 13m 56s","accounts_in_message":[],"_revision_number":1},{"id":"6a557503d7abdf8444c1c9bf47dfc98bacc9582e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-16 21:31:23.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1655387687998 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"dfc36fd7bb020782154b5e42f17d4db0d5cd759a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-16 23:18:34.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-73-846173-1/ns5_nfs : SUCCESS in 3h 39m 5s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-73-846173-1/ns5_iscsi : SUCCESS in 3h 45m 35s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"5ebf0137be6cae4a4dfe2669f40d5355fb0ceb27","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-06-17 00:48:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/846173 : FAILURE in 1h 19m 41s","accounts_in_message":[],"_revision_number":1},{"id":"79f6cd6ee732289909a325a5dae135afc1f3a13b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-17 00:52:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k-focal https://hpe-storage.github.io/alletra-6k-ci-logs/logs/73/846173/1/check/iscsi-driver-alletra-6k-focal/9516cc5/9516cc5.html : FAILURE in 51m 25s","accounts_in_message":[],"_revision_number":1},{"id":"39c2e4d8fba5ec9b2fa7a3cdc3ddbd01e1ed1a4b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-17 01:05:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1982d1005c134c348b07be2af21a6d8a : SUCCESS in 39m 31s","accounts_in_message":[],"_revision_number":1},{"id":"626da58b72d24b4900e6935c19e0900f777b5798","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-17 05:19:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/73/846173/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b2399c0/ : TIMED_OUT in 3h 03m 00s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/73/846173/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/1f26649/ : FAILURE in 2h 54m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f57d82fe3fbfed764d19ae988e6b00e8d0b16b24","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-06-17 08:01:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/73/846173/1/ : RETRY_LIMIT in 8s","accounts_in_message":[],"_revision_number":1},{"id":"833ea4d583576cd2debf83972a8a49555fc9e225","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-17 09:53:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/73/846173/1/check/iscsi-driver-nimble/910edad/910edad.html : FAILURE in 48m 41s","accounts_in_message":[],"_revision_number":1},{"id":"29efdd350dba1f5f2a08b32c05f2233a4783ee65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-06-17 10:50:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"89af2ab5b95b194e94b9f888317a69f14ce36c06","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-17 11:08:32.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1655463031724 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"ded29519a304d63b863037b8ae12fefa4da93248","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 11:27:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/25c7f09554b648ffabe5675ed4087a38 : SUCCESS in 21m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/659c365cc4ef490ca3a9c83f9f4a8674 : SUCCESS in 21m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3a29fc271dfb69e13dbf62adc4f35f718eed7f27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 13:29:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a9707b57a924c8793c371d0aafe75bb : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/477dd9e9e0424f5dae434e455f0ac6dc : FAILURE in 11m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eba24c1f1c7b4a68a7e85ee4a7649a52 : SUCCESS in 12m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e7fcd896e564913b1168060b02a0f9e : SUCCESS in 12m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2802d4e7af347c6ad1fd4c5a5d74771 : SUCCESS in 11m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/2d536f2e8a634536bc68773323bcc616 : SUCCESS in 59m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4e46cb22de954adfa3aace537a5fb474 : SUCCESS in 1h 19m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02053ab070b34de184d02f745095a05a : SUCCESS in 54m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/278b482632fe46548bf7c2c6f4358108 : SUCCESS in 16m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1d8d3375baaa42ae9698efc3d5ca6a0f : SUCCESS in 6m 44s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fb9b3ddfa2584a579076a66b22dddbe4 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/95df3ca8d9734339b2945bd699cd0c7e : SUCCESS in 6m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5f7ff294f05541cb98fac4f4136e71cd : SUCCESS in 8m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4767bf184aed48409112c528ae8103a3 : SUCCESS in 35m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8071cff810b54584b3ad0ca99ec50104 : FAILURE in 7m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/43b96b2e54f848bd93f33d63329428a2 : SUCCESS in 1h 17m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fd87f6cc833949a4ad46e43d91431c2d : FAILURE in 1h 14m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4418d79757234146a66d4cb00b5f157c : SUCCESS in 1h 40m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/592b14a9b2f745acad00dedc64648f0f : SUCCESS in 1h 21m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/50bfeeb025fc496ca580fc4b2160ec1a : SUCCESS in 1h 26m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ca4ce49496234b7f955114568992c03b : SUCCESS in 44m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/56b71d572660471bbca02e2978c02b09 : FAILURE in 1h 14m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9e0a23b17c7946dfabc8ee943ab83807 : SUCCESS in 2h 27m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55787a71fe524a75b2578331506096db : SUCCESS in 50m 51s","accounts_in_message":[],"_revision_number":2},{"id":"198e294ed93def437dc6276240c743b602b607a9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-17 14:04:16.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/846173/2/iscsi/ : SUCCESS in 57m 37s\n- cinder-smb http://cloudbase-ci.com/cinder/846173/2/smb/ : SUCCESS in 1h 05m 34s","accounts_in_message":[],"_revision_number":2},{"id":"811645447d88402a3491ca5d8c0fc8882f923930","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-06-17 15:46:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a412b78831476b0a35763fe32dcf99263ce704cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 16:13:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f566b38c332c4e5b8c58a23758403c05 : SUCCESS in 23m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d8562f2bcfc14cf2b2e157da23d061f2 : SUCCESS in 15m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"279d7a5a872334e38ccf6999401d1dc45d56cf9c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-17 16:18:25.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/5544eba4450143ccb2ee705793ef3720 : POST_FAILURE in 2h 23m 54s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/0f264f28a96d4f8a84e0058d826e2422 : TIMED_OUT in 3h 05m 50s","accounts_in_message":[],"_revision_number":2},{"id":"fd97891bf33bafb1585f5817d10716f58877c546","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-17 16:23:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k-focal https://hpe-storage.github.io/alletra-6k-ci-logs/logs/73/846173/2/check/iscsi-driver-alletra-6k-focal/5ef65a0/5ef65a0.html : FAILURE in 50m 21s","accounts_in_message":[],"_revision_number":2},{"id":"ddddba4763a6734c33ee0385e06db7999a611694","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-17 17:02:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/73/846173/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f9696c7/ : TIMED_OUT in 3h 03m 01s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/73/846173/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f9cd01b/ : FAILURE in 1h 34m 53s","accounts_in_message":[],"_revision_number":2},{"id":"8a44a8a4804429a5e13de4cdfa43641fe55a5028","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-17 17:03:29.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-73-846173-2/ns5_nfs : SUCCESS in 3h 37m 26s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-73-846173-2/ns5_iscsi : SUCCESS in 3h 43m 27s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"772023e30f03d1eae46fb237e83d31b98ed9ef9f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-17 17:14:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/846173/2/iscsi/ : SUCCESS in 1h 24m 17s\n- cinder-smb http://cloudbase-ci.com/cinder/846173/2/smb/ : SUCCESS in 1h 11m 03s","accounts_in_message":[],"_revision_number":2},{"id":"ce1fe20ff2e8dcfd7475b9278cd4159b0dcc5948","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 17:53:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab3d5299a5e14b8dad4f9b017a28137a : SUCCESS in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f93eda6fadf34d6687481926ec839667 : SUCCESS in 12m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/980ab1cfdc3a478394dafaab94124850 : SUCCESS in 11m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dff08e2d02334687aa08576eb6b81585 : SUCCESS in 11m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81118232e73a4ddd9504fd3486bf34f1 : SUCCESS in 15m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/0f28eda2c32840bb895a246b2d7dac8d : FAILURE in 39m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a5129e9a53c342e685fda2d6cffc4b1e : SUCCESS in 1h 19m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f8d967955a940a9a968f12bbef96b7a : SUCCESS in 49m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/948139877d154857ad90b089d02edc38 : SUCCESS in 15m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f4ae9feadc2f4c96a4083ccb11e30174 : SUCCESS in 5m 28s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/559a09dedcea48d99fc5a6a4a454b456 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/193250e44e204115971717ca49c243be : SUCCESS in 5m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4f8e648086124f959745716663a7f78f : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7b565eabcf3347f1a209a3a6201c3121 : SUCCESS in 38m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b04fd95730894e8bad787c8f16fe92bd : FAILURE in 7m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/69a0e964c76c4199a49e25f92d99fa07 : SUCCESS in 1h 17m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7c42a6615e3f48d5938286ef770bf54f : FAILURE in 1h 17m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f0c31993867347238b688c5868d634f2 : SUCCESS in 1h 26m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cf8fba40b4e1407a8ed8d60e4715a9c2 : SUCCESS in 1h 00m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8504770db1f347b2bde9aa3da34f0fa7 : SUCCESS in 1h 14m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c933ba502ef34f26903d579703ec8000 : SUCCESS in 32m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0c41edc5331343d38cfbb145234637fe : FAILURE in 1h 00m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1bce9078288740fbb6dbb273af94c0d1 : SUCCESS in 2h 02m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfc083661a2040e3acb8db531e3463fc : SUCCESS in 47m 30s","accounts_in_message":[],"_revision_number":2},{"id":"b36488bafe6c8c19b5acce955e022944777d6438","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-17 18:23:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b2f519308d20404282dba0050beec837 : SUCCESS in 41m 49s","accounts_in_message":[],"_revision_number":2},{"id":"b52c002f1addf4d509caba05bdbdd4f7a05487e4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-17 21:58:12.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/73/846173/2/check/iscsi-driver-nimble/04df2ba/04df2ba.html : FAILURE in 48m 19s","accounts_in_message":[],"_revision_number":2},{"id":"aef7ba8ad5ce345b98d367906057617b7d95c5a8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-18 12:26:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/73/846173/2/check/dsvm-tempest-cisco-zonemanager-job/353f0e7/ : SUCCESS in 1h 07m 07s","accounts_in_message":[],"_revision_number":2},{"id":"b9ecff2f6dd1672f3b61df93a7cd55e9cde1bbee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-20 05:26:53.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"aa8f60c8bc6509134b66a27605fa2d01d8f7d4ff","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-20 09:32:09.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-cinder http://207.189.188.190/logs/846173/846173/2/check-not-vote/ext-cinder-zuul/e857094 : FAILURE in 2m 17s","accounts_in_message":[],"_revision_number":2},{"id":"7030a6eec37024e851df35891483282d8efb4438","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-06-23 21:23:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"733be066c2c9f93f0b78061ec535853f532b7dbf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 21:45:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d67bc53ac914ecd9af510b113430285 : SUCCESS in 14m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c03a2082db0a492ab14cf374cb5b838e : SUCCESS in 15m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"68d7086ca6dc0f7ce12ab3ee9bb2e1de6b7d75c3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-23 22:56:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/846173/2/iscsi/ : SUCCESS in 1h 18m 07s\n- cinder-smb http://cloudbase-ci.com/cinder/846173/2/smb/ : SUCCESS in 1h 25m 09s","accounts_in_message":[],"_revision_number":2},{"id":"7d0328d6d21d9b32ffebdf3f0f79b56b4d8e0792","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-23 23:14:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/661dd88a43df4ec8bdf1e596b567331e : SUCCESS in 39m 59s","accounts_in_message":[],"_revision_number":2},{"id":"2826dcb827564d71f39e73b5556fb63ab399cb4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 23:33:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/159604e19ce04f709d0f20a79af96501 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa51f3ae1259423ba6d8529010330392 : SUCCESS in 9m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d383816ab624cca8cc79b7295283de1 : SUCCESS in 11m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be10e2aa78974f98ab5c21b5f8c955bb : SUCCESS in 12m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa23bd8aef1a4d6fab0c83ccb3a7a2db : SUCCESS in 11m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/1edaf971df2140fbae5323a2e6849604 : SUCCESS in 44m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f3c7a98e5da2430182c0d1b360b2c39a : SUCCESS in 1h 05m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c604bb612fcb4856b492cd42db6247b5 : SUCCESS in 46m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9cd19271f32a496ea87f2d093506b26c : SUCCESS in 14m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a4d8e42acc454ac199d8f23e0ef90b1f : SUCCESS in 5m 30s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cc676b38842b4614a6b373558856620c : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b9adf75d11574683819bde37257360f6 : SUCCESS in 6m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1b9ff91ab93f42068284ed69117f3a97 : SUCCESS in 7m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/33a141ebb2b64e6791a295cbe3a08008 : SUCCESS in 29m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/19d3cb6744ea4ec2b74e1967d14d6888 : FAILURE in 6m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b4b3ca207b4340daa5fa59b70316d358 : SUCCESS in 1h 14m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e8ecd09d80a843a48a2178f7e8c83b7d : FAILURE in 1h 24m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6fdf21147bc94ecfb45f4f473d82ab79 : SUCCESS in 1h 43m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a327742caf6c43919e948614fd4d46d3 : SUCCESS in 57m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/de14c738a52b4c08a92f1490f16a9d3f : SUCCESS in 1h 18m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b96d08caf96b44cda4efe77b24286279 : SUCCESS in 29m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/87f21c1956ba4905abf3776466f227fd : FAILURE in 1h 03m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/32751d2def624a67a591d9c830709cff : SUCCESS in 2h 07m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e8c6b58e4524808a93bce7d10aab514 : SUCCESS in 50m 17s","accounts_in_message":[],"_revision_number":2},{"id":"c416d29dbc02b3eacea57e7aba95b91c4b3dc115","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-06-23 23:34:36.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"dd1f19f03d160c0fece5082ab6fb6143424c9b11","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-06-23 23:50:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-73-846173-2/1656027187 : SUCCESS in 17m 20s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"6cee5e3e1f6b421135b86c640438cd9a6ccd1482","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-24 00:37:38.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/5de2cf6ae8a24af6a405b2d7cc69e04b : POST_FAILURE in 1h 43m 59s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/17ef436ffb2847f59219aa67e51df093 : TIMED_OUT in 3h 08m 20s","accounts_in_message":[],"_revision_number":2},{"id":"6e4981a52e243ed5a48a092514e92536a28f2145","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-24 00:56:48.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/846173/2/upstream-check/cinder-tempest-plugin-ontap-fcp/cf5c398/ : SUCCESS in 1h 11m 52s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/846173/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/3db43f1/ : FAILURE in 1h 20m 47s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/846173/2/upstream-check/cinder-tempest-plugin-ontap-nfs/e7cb95a/ : SUCCESS in 41m 37s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"55944d18c55ad94fe1259dae9b1aabd7e5ef6c5c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-06-24 00:59:17.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/73/846173/2/check/DellEMC_PowerFlex/534d5f2/DellEMC_PowerFlex/899 : SUCCESS in 1h 25m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"fddf3e8895111a22b01df66dd1a2bbb3cd613d30","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-06-24 01:24:53.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/73/846173/2/check/DellEMC_SC_FC/75bd3ad/DellEMC_SC_FC/745 : SUCCESS in 51m 27s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/73/846173/2/check/DellEMC_SC_iSCSI/1caa969/DellEMC_SC_iSCSI/767 : SUCCESS in 1h 50m 57s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"39388eb86d76d74a7929fb0214a43e4a552d64e6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-24 01:48:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/73/846173/2/check/dsvm-tempest-cisco-zonemanager-job/c278048/ : SUCCESS in 1h 10m 22s","accounts_in_message":[],"_revision_number":2},{"id":"e0610d6bd777fdcd576414a3de3cf14fcb9984d8","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-06-24 03:44:46.000000000","message":"Patch Set 2:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/846173-2/HPMSA-iSCSI/ : FAILURE in 54:28.22\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/846173-2/HPMSA-FC/ : FAILURE in 51:37.24","accounts_in_message":[],"_revision_number":2},{"id":"13d491faead9677644bc9bf393e93b1752876ca6","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-06-24 13:44:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/846173-2/Lenovo-iSCSI/ : FAILURE in 1:11:13\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/846173-2/Lenovo-FC/ : FAILURE in 1:06:14","accounts_in_message":[],"_revision_number":2},{"id":"4fa62eec432a1164bcf8ca4fcbc49ec5294770b9","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-06-24 17:15:45.000000000","message":"Patch Set 2:\n\nBuild failed.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/846173-2/Seagate-iSCSI/ : FAILURE in 1:10:04\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/846173-2/Seagate-FC/ : FAILURE in 1:10:17","accounts_in_message":[],"_revision_number":2},{"id":"1640a7e21f27efe75f32c9dab7ce3a009f0e02a3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-24 18:35:23.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/73/846173/2/check/DellEMC_Unity_FC/fbef71f/DellEMC_Unity_FC/1020 : SUCCESS in 1h 04m 37s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/73/846173/2/check/DellEMC_Unity_iSCSI/ecfc98d/DellEMC_Unity_iSCSI/956 : SUCCESS in 1h 50m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"dd487e4f41d150a8d3f28a66d6d3f51b249732ea","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-06-26 16:45:36.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/73/846173/2/check/DellEMC_PowerMAX_FC/70583c1/DellEMC_PowerMAX_FC/618 : SUCCESS in 3h 14m 51s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/73/846173/2/check/DellEMC_PowerMAX_iSCSI/ceadb81/DellEMC_PowerMAX_iSCSI/605 : ABORTED in 10h 30m 01s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"d68c7ee7624b3abb4510c8004f359d217605fc0d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-13 12:33:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/2/20220713144448/NEC-V-FC : SUCCESS in 1h 53m 04s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/2/20220713144448/NEC-V-ISCSI : SUCCESS in 6h 48m 38s","accounts_in_message":[],"_revision_number":2},{"id":"7f8387f4ce91a14e8d14932a0381e3775e7e1834","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-07-16 09:30:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/2/20220716150025/NEC-FC : SUCCESS in 3h 26m 19s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/2/20220716150025/NEC-ISCSI : SUCCESS in 3h 29m 50s","accounts_in_message":[],"_revision_number":2},{"id":"9803ce9059d63b77792b65c8a1025843cf0e5579","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-18 15:45:39.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"b8b13ee4dae4023ece2cb9d747c21c4a0f7d3f69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 16:09:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/852aa5968b104afd8257fa28e831b4bb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2808e362972d40229609a0c604895d9e : SUCCESS in 14m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1338af9ec5ac49ecafaacae823a7ff05 : SUCCESS in 20m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"89af865f3b7ed830958db94e828c00dc3a680bf5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-18 17:06:55.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/846173/3/iscsi/ : FAILURE in 1h 17m 03s\n- cinder-smb http://cloudbase-ci.com/cinder/846173/3/smb/ : SUCCESS in 1h 19m 17s","accounts_in_message":[],"_revision_number":3},{"id":"9acce2ba53c4baa4df97d17083e964fa819611eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 18:21:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf3b4315c4254b2bb8e0fa6192868296\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78564001f03a4047ba2ddf0ec5068232 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/386028f094eb4a70a460159746debbac : SUCCESS in 11m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61e30ba4b3e54a76af88eedbd04bd9e8 : SUCCESS in 14m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8a12c112ae6463ea085c4cee34e3196 : SUCCESS in 14m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4652ad59f7ec4b98b01a0b07fae77599 : SUCCESS in 11m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/f67036be86ea4081800e9f11788ffbdc : SUCCESS in 51m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c889b1fed88644a0b2f1f366d2a4324b : SUCCESS in 1h 29m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74b9a21ac82b460bba82d07cbf3bc51c : SUCCESS in 1h 01m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c9349cca5478405da53e5bc264b1331e : SUCCESS in 14m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2363a77134f8482899edd5bd226a86d0 : SUCCESS in 7m 43s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0872c6d1da104c4591ac2ab2e267aaf3 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7343b1ca1c954167a5f1ec814bbc78ea : SUCCESS in 8m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c729193960b843c8a121f0f94bc452e0 : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a264f61db0114e46acbf1753dcb4bad7 : SUCCESS in 42m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dcc3a9bdb7164fbca2e9fe6b56d4baef : FAILURE in 8m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1fa99cc33be4414d8e3a50361eaba1be : SUCCESS in 1h 08m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/712bd92f658c4dcdae7a9eddc9781fab : FAILURE in 1h 44m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cabeb2d41fb8412e83530117f6352fc0 : SUCCESS in 1h 41m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/364bf13cbb524ab2a387b479f8b533be : SUCCESS in 1h 20m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/38e8485280ae496090dc99c94804fc4f : SUCCESS in 1h 19m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fb3b69f9633544d094059bdfec8d1c9f : SUCCESS in 52m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9abea4d72f484e53baaa71c7bb4c5b09 : SUCCESS in 1h 23m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/161ec6d8c71e449dbc251085f67ccff1 : SUCCESS in 2h 28m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea510d3e4c5743d9979d9b001deaeb45 : SUCCESS in 55m 53s","accounts_in_message":[],"_revision_number":3},{"id":"80719f5aebfd5e1eca272bfcc4620dbe9d705a56","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-18 18:30:22.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/846173/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/2f16555/ : RETRY_LIMIT in 7m 28s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/846173/3/upstream-check/cinder-tempest-plugin-ontap-nfs/59dd562/ : RETRY_LIMIT in 7m 28s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"254472872e3313b51bcaefab86e77a7834149983","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-18 18:34:31.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/846173 : FAILURE in 1h 00m 18s","accounts_in_message":[],"_revision_number":3},{"id":"a2f538c7c5d1e825983061ee39f899211a3cac24","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-18 19:01:46.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1658159147548 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"3e82e38f6d284299354de0e6299131c36e51b588","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-18 19:17:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/73/846173/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b475cc8/ : SUCCESS in 1h 43m 35s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/73/846173/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/4494ad9/ : SUCCESS in 1h 40m 16s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/73/846173/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/fc90d6a/ : FAILURE in 28m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"32aef8437c9c5052b9ae06dcc654a01f804977ac","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-18 20:21:57.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-73-846173-3/ns5_nfs : SUCCESS in 3h 24m 13s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-73-846173-3/ns5_iscsi : SUCCESS in 3h 28m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"d2c0090c35a26fe0a47fb715b3d41d0080d86af5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-18 21:51:05.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/2afae9c2ad4a400b855c30b74bbfe321 : RETRY_LIMIT in 1m 00s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/73/846173/3/check/devstack-plugin-ceph-dsvm-focal-py3/274fbe1/ : TIMED_OUT in 3h 14m 47s","accounts_in_message":[],"_revision_number":3},{"id":"84d4e4cc8b3c0605b4014ff619c1a50f8e330373","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-18 23:20:31.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/73/846173/3/check/dsvm-tempest-cisco-zonemanager-job/757d74b/ : FAILURE in 1h 35m 08s","accounts_in_message":[],"_revision_number":3},{"id":"d3092e3c5c2e45760e70f3f7ed9acd726bf4ac89","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-18 23:37:07.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/73/846173/3/check/DellEMC_PowerStore_NFS/0fa9236/DellEMC_PowerStore_NFS/144 : FAILURE in 41m 50s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"d6e0f7b140f8eace944fa69a7a3f929a2badb383","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-07-19 00:18:33.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/73/846173/3/check/DellEMC_PowerMAX_FC/5b81d09/DellEMC_PowerMAX_FC/730 : ABORTED in 30s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/73/846173/3/check/DellEMC_PowerMAX_iSCSI/48d6079/DellEMC_PowerMAX_iSCSI/717 : ABORTED in 1m 42s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"ac07a27564aa404c0f5a11d7711785de3586d00a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-19 01:26:41.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/73/846173/3/check/DellEMC_PowerFlex/42f3891/DellEMC_PowerFlex/1017 : SUCCESS in 1h 27m 52s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"8adf6257c7f738d50fd9de3451e063a51fcc74a9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-19 04:41:31.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/73/846173/3/check/DellEMC_Unity_FC/fa3f9dd/DellEMC_Unity_FC/1182 : SUCCESS in 1h 13m 22s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/73/846173/3/check/DellEMC_Unity_iSCSI/0854885/DellEMC_Unity_iSCSI/1106 : SUCCESS in 1h 55m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"be7bf94edb4166a91fa36c1da2df7a8984e8faf7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-23 23:11:21.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/3/20220724055036/NEC-V-FC : SUCCESS in 1h 57m 20s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/3/20220724055036/NEC-V-ISCSI : SUCCESS in 2h 20m 42s","accounts_in_message":[],"_revision_number":3},{"id":"04c5db7447ad2b96f48af6746254ab0b6be0bcc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-28 16:44:21.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"2324cf1eac1e552ed81475cc4dc3fd881e5ba152","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-28 16:44:25.000000000","message":"Patch Set 4:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":4},{"id":"af470cfbee4a6759b05346f895d3d18114ac7f6c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-28 16:50:11.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"d0a662a0cef1fa17cf002b2dd287180411c31a17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 17:13:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8590a2052574a4d9c4ba96f648a0842\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/169e0c871293448895acabce4a7d1eec : SUCCESS in 18m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/68cf6bbbb83140dfb50005b61583e543 : SUCCESS in 17m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d029d7a81702071f99946c0b2f2771932128a8cd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-28 17:27:28.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1659026671684 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"ef814aef55a26c38ee2c9473f2c82c0681aefdad","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-28 18:33:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/846173/4/iscsi/ : SUCCESS in 1h 47m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/846173/4/smb/ : SUCCESS in 1h 34m 04s","accounts_in_message":[],"_revision_number":4},{"id":"2c771e4f8aafea11c8d9ce7a23d035feab5d4be6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-28 18:35:12.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/73/846173/4/check/iscsi-driver-alletra-6k/d30d116/d30d116.html : FAILURE in 52m 47s","accounts_in_message":[],"_revision_number":4},{"id":"4948a2f9f1c78d80cea53b054f60ca130b56ecd2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-28 19:03:35.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/7a7bf32faa8b41ecb9bc26c40210b3cf : POST_FAILURE in 1h 41m 01s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/ae5c211efd6a4fa0a8f1b689c8122964 : POST_FAILURE in 1h 50m 44s","accounts_in_message":[],"_revision_number":4},{"id":"b01f353ed8028bfa5a461495b3852de96bd92530","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-28 19:04:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/846173 : FAILURE in 59m 59s","accounts_in_message":[],"_revision_number":4},{"id":"5fef6a716f887899df5e8d70f30c27368296bb02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 19:07:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ba6cc32203e49b9b98e865a764ab8b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90169c06eaf34029a7be7f5f8daf8092 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/956cb885620946d9aa2785331b07366a : SUCCESS in 12m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29d8a293f58142b78555fcbf710498d8 : SUCCESS in 12m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef0dd3295763464a84672887b457a50f : SUCCESS in 14m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d66bad328cba4caa81c8f106d6d7cd44 : SUCCESS in 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/f41ab0094c2a4afeb1e7c0a333ce17db : SUCCESS in 1h 16m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/91d5fef7de764b428d719a9d872f65a1 : SUCCESS in 1h 35m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eef627f2e6b249a8a722b1da192bee8e : SUCCESS in 53m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a4371165a26a42bfb59d69076a4468da : SUCCESS in 17m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8360cfebebc6484694d3143dc0bfffa8 : SUCCESS in 5m 58s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/efcffcb4331441da897f0db51750a460 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2cbfe7ce91684f90a83961a24ad211ff : SUCCESS in 6m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/65faf5006a6b4f0ba667532db186a8c2 : SUCCESS in 6m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f6ee12f9dee04beeae0ea15b1450ce3d : SUCCESS in 46m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/78c6bb556b8541e19c284560fb0a87e2 : FAILURE in 6m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9450f76817e9452e940e761684efe1ea : SUCCESS in 1h 20m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/835e5c17d21346a28b27aa261f69e3b0 : FAILURE in 1h 40m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2a29f8db740c4fceaa0a341ce5a57612 : SUCCESS in 1h 58m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/99930506e8f34628870e655298c4679b : SUCCESS in 1h 25m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/475c29bdc8274b21b052f6e934a62c28 : SUCCESS in 1h 38m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6b639f17f1be41d08176513334c49b66 : SUCCESS in 48m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ba44da7f856e4999add9b673a1eefe72 : SUCCESS in 1h 25m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e74657d8d44c481fb286ea23d35e01ba : SUCCESS in 1h 58m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4709b0c993d8468aacb3531a5b689fc6 : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":4},{"id":"9074ddf692e312ede34f76aae3814adbaffafea7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-07-28 19:15:49.000000000","message":"Patch Set 4:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/28_07_2022_19_07_25\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"ef01694e45508d546c99839625a613e040c6f014","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-28 19:51:39.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/73/846173/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/9b6cffb/ : SUCCESS in 1h 33m 26s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/73/846173/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/cdc5621/ : SUCCESS in 1h 35m 08s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/73/846173/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/a51c5ad/ : FAILURE in 29m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a41cab49d687c02c27d2f0d5bda39615b3756051","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-28 20:37:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/846173/4/upstream-check/cinder-tempest-plugin-ontap-fcp/46b2825/ : RETRY_LIMIT in 8m 36s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/846173/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/ae16ec7/ : RETRY_LIMIT in 8m 00s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/846173/4/upstream-check/cinder-tempest-plugin-ontap-nfs/b8fa03c/ : SUCCESS in 1h 14m 22s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/846173/4/upstream-check/cinder-tempest-plugin-solidfire-iscsi/92962c1/ : SUCCESS in 1h 28m 45s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"d533466bebc21717e49c94d52def02d5854781f5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-28 21:17:08.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-73-846173-4/ns5_nfs : SUCCESS in 3h 28m 2s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-73-846173-4/ns5_iscsi : SUCCESS in 3h 26m 16s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"44ac53a15552b1c3a3cf517ad7f4043dbcb50900","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-07-28 21:35:51.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/73/846173/4/check/DellEMC_PowerMAX_FC/823c72b/DellEMC_PowerMAX_FC/775 : FAILURE in 42m 01s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/73/846173/4/check/DellEMC_PowerMAX_iSCSI/0c80543/DellEMC_PowerMAX_iSCSI/758 : FAILURE in 50m 00s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"f06e48c17d1ee98885300649a80072a4fc3561d3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-28 23:45:08.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/73/846173/4/check/DellEMC_PowerFlex/531ba06/DellEMC_PowerFlex/1101 : FAILURE in 1h 33m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"7b786beadacca2a7306ad483dfaf0efef2f8ec03","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-29 01:28:30.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/73/846173/4/check/DellEMC_PowerStore_NVMeTCP/8736ba1/DellEMC_PowerStore_NVMeTCP/55 : FAILURE in 1h 49m 54s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"149a007e32713ca8aea04e7c19308346820adc1d","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-07-29 18:18:49.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/73/846173/4/check/DellEMC_SC_FC/042ee08/DellEMC_SC_FC/832 : SUCCESS in 55m 00s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/73/846173/4/check/DellEMC_SC_iSCSI/95418ba/DellEMC_SC_iSCSI/851 : SUCCESS in 1h 54m 49s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"544286e4a047ba9dc2e5854dc3f3ddaabfea6c16","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-29 22:40:24.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/73/846173/4/check/DellEMC_Unity_FC/16fdb50/DellEMC_Unity_FC/1302 : FAILURE in 23m 28s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/73/846173/4/check/DellEMC_Unity_iSCSI/e8630ae/DellEMC_Unity_iSCSI/1184 : SUCCESS in 1h 52m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"289d8b2b3a82612d81e2979cec2cc7b732c55d6e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-30 18:17:54.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/4/20220731005755/NEC-V-FC : SUCCESS in 1h 59m 05s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/4/20220731005755/NEC-V-ISCSI : SUCCESS in 2h 19m 57s","accounts_in_message":[],"_revision_number":4},{"id":"b3db7b06725cb57675d22b54bd878014216394f1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-06 21:27:12.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/4/20220807025415/NEC-ISCSI : SUCCESS in 3h 32m 07s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/4/20220807025415/NEC-FC : SUCCESS in 3h 32m 54s","accounts_in_message":[],"_revision_number":4},{"id":"cb5bf948e4d6b9fe39b06593fc9e3410dcb72e80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-08-31 08:56:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1efcac49b80fdcb47e8d564512366dbf115370d2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-31 08:56:48.000000000","message":"Patch Set 5:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":5},{"id":"dec90dfdd06beb253127fbaa5d41a49b364f7df2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 09:29:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaa0bf3f32fb4f87ad869f5fee27743e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d7c11e4b8b74fd4a1ee3b5429188acf : SUCCESS in 21m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0fefa5dfa7e94f2681487dc531ede040 : FAILURE in 25m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c977235647bdae72a90baa1a0252d2fbdc4e5310","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 11:04:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/052b399a962a40deb0520d3ac51a7df7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c9415b6a8464c819b7f285de7c0baa4 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/916813d0035440878bd1eb33cbea4cdc : SUCCESS in 13m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2870d6c58d69490a8f91ab3411c2d237 : SUCCESS in 10m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00a8bf36b9fd4ed7b7a3d71b0a313420 : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/160b5bfd376847acaed9596b9f013002 : SUCCESS in 11m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/a5ae7e0ecd7d43ceadf3a66268ef092d : SUCCESS in 51m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/682d9a7bf29c4ff7883b96e67f02bca3 : SUCCESS in 1h 25m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/712fcda91dd7457c92ca985161c34bbc : SUCCESS in 1h 06m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/aecac27c824e4ac8aaa9066129d2228b : SUCCESS in 16m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/63a961676bc3407191f4d964c1d2d191 : SUCCESS in 5m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a3a0d1126f2c402699bfd1f38549c2fe : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/44e466f229aa42289211b296ebb82d03 : SUCCESS in 6m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8bee195adcd04fe594c7152c9ccea062 : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3b6ec6a6c3c5471b95672f952bdb7355 : SUCCESS in 45m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6c12015f1c8841c29cd7deb87ac5138b : FAILURE in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3ad5de560b7f4739a282e6a95dea241b : SUCCESS in 1h 29m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/75cc46b06a1b4c328d9b8e30f4cc3e0b : FAILURE in 1h 30m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b0541b102e0f427e92e5fdefb73fcae5 : SUCCESS in 1h 56m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b5ebed5c88924347ae6aac891b3267ef : SUCCESS in 1h 11m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0b139e590a2b4230a56dccea8dfd433c : SUCCESS in 1h 43m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fa7a6f14f1484ef3b124a1dc6aef6053 : SUCCESS in 43m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f04c7ee7c0eb4caa9c3759ccd9530933 : SUCCESS in 1h 29m 30s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/972929f5f62443f39894a779b638d20e : SUCCESS in 2h 04m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b86de4727bdf45a8996b13a0b05fb2b2 : SUCCESS in 1h 00m 25s","accounts_in_message":[],"_revision_number":5},{"id":"ef8a11ad403c156dd69c354cbb7e2773de419334","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-31 11:19:35.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_08_2022_11_11_39\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"28206bd771c5f8dbf8df33497534e008d64f1019","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-31 11:52:11.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/846173/5/iscsi/ : FAILURE in 1h 49m 41s\n- cinder-smb http://cloudbase-ci.com/cinder/846173/5/smb/ : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":5},{"id":"27daab6a74e9b1836e0bd1fa0ab5cf327c0d17be","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-31 12:15:49.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/846173/5/upstream-check/cinder-tempest-plugin-ontap-fcp/2d18a31/ : SUCCESS in 1h 07m 26s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/846173/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/d1ef51b/ : SUCCESS in 1h 04m 27s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/846173/5/upstream-check/cinder-tempest-plugin-ontap-nfs/e14a802/ : SUCCESS in 43m 03s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/846173/5/upstream-check/cinder-tempest-plugin-solidfire-iscsi/a6d7318/ : SUCCESS in 51m 58s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"cd12fc4fc9069fbc68ff9f7a1d8f06245d287586","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-31 12:39:38.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/73/846173/5/check/iscsi-driver-alletra-6k/6b364a5/6b364a5.html : FAILURE in 52m 03s","accounts_in_message":[],"_revision_number":5},{"id":"1c1696072b728d82a1ea30714cef8eea0d88d8df","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-31 12:45:20.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/846173 : FAILURE in 1h 15m 47s","accounts_in_message":[],"_revision_number":5},{"id":"1d1aa8a7c775422b92919a808c22931543950344","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-31 12:57:37.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/846173 : FAILURE in 1h 22m 59s","accounts_in_message":[],"_revision_number":5},{"id":"395911071efce611174c79eb26fec996e9b0f799","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-31 14:04:47.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-73-846173-5/ns5_nfs : SUCCESS in 3h 32m 55s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-73-846173-5/ns5_iscsi : SUCCESS in 3h 37m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"6d9e938c0ed67b70c333826e1fbbcb0f40b5c1ad","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-31 14:07:52.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/73/846173/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2b7ec44/ : SUCCESS in 1h 49m 54s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/73/846173/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6027cd1/ : SUCCESS in 1h 34m 56s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/73/846173/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/ba92846/ : SUCCESS in 1h 42m 18s","accounts_in_message":[],"_revision_number":5},{"id":"62a205716963d43728e1a665bb6114977d8f786d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-31 14:30:16.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/73/846173/5/check/dsvm-tempest-cisco-zonemanager-job/3aaf670/ : SUCCESS in 1h 07m 08s","accounts_in_message":[],"_revision_number":5},{"id":"407210b3880e0412bc2e237b65eacacee6863c02","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-31 15:10:36.000000000","message":"Patch Set 5:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/73/846173/5/check/fujitsu-eternus-dx-iscsi/b34f7ed : FAILURE in 2h 25m 37s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/73/846173/5/check/fujitsu-eternus-dx-fc/fa5fb2f : SUCCESS in 1h 47m 00s","accounts_in_message":[],"_revision_number":5},{"id":"d5d35d897310330f258297aa6155bddcd2465ffa","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-31 19:25:47.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/73/846173/5/check/DellEMC_PowerFlex/11d5a97/DellEMC_PowerFlex/1260 : ABORTED in 2h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"73b3413efcefce50272667ff64881f99e28da5cd","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-31 20:45:38.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/73/846173/5/check/DellEMC_PowerStore_iSCSI/c8480dc/DellEMC_PowerStore_iSCSI/676 : SUCCESS in 1h 53m 52s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/73/846173/5/check/DellEMC_PowerStore_FC/447eedd/DellEMC_PowerStore_FC/831 : SUCCESS in 1h 17m 47s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/73/846173/5/check/DellEMC_PowerStore_NFS/6996488/DellEMC_PowerStore_NFS/367 : ABORTED in 3h 00m 01s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/73/846173/5/check/DellEMC_PowerStore_NVMeTCP/b71ec33/DellEMC_PowerStore_NVMeTCP/241 : FAILURE in 1h 46m 18s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"e49f53d5effdcf10036f3517d1dbeeb13a97c208","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-09-01 06:28:55.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/73/846173/5/ : SUCCESS in 1h 28m 13s","accounts_in_message":[],"_revision_number":5},{"id":"8432538fe8c398cf1a1a5152fa356f2358eb5a52","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-09-01 12:29:31.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/846173-5/HPMSA-iSCSI/ : SUCCESS in 1:03:00\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/846173-5/HPMSA-FC/ : SUCCESS in 1:01:31","accounts_in_message":[],"_revision_number":5},{"id":"21568f723497b1ea5f69cb4bc5fdf30e503bbabb","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-09-01 21:05:02.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/846173-5/PVME-iSCSI/ : SUCCESS in 1:06:13\n- PVME-FC https://os-logs.seagate.com/dell-ci/846173-5/PVME-FC/ : SUCCESS in 1:04:43","accounts_in_message":[],"_revision_number":5},{"id":"f724be4a3dabd56c0743df3d2db84299b3953d43","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-09-01 21:05:40.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/846173-5/Lenovo-iSCSI/ : SUCCESS in 1:06:25\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/846173-5/Lenovo-FC/ : SUCCESS in 1:06:36","accounts_in_message":[],"_revision_number":5},{"id":"0b065dfdfa099f66d26439e97c620859869c0c05","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-02 03:11:06.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/73/846173/5/check/DellEMC_Unity_FC/d850d84/DellEMC_Unity_FC/1644 : FAILURE in 1h 20m 53s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/73/846173/5/check/DellEMC_Unity_iSCSI/f6180ad/DellEMC_Unity_iSCSI/1501 : FAILURE in 1s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"fd2a5e0faae7091406d1a8cd1d233dae512d14bc","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-09-02 05:10:33.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/846173-5/Seagate-iSCSI/ : SUCCESS in 1:14:53\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/846173-5/Seagate-FC/ : SUCCESS in 1:14:03","accounts_in_message":[],"_revision_number":5},{"id":"0068a8a5055ba796abfa7108891785035e9e83f2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-09-07 17:13:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/5/20220907223746/NEC-ISCSI : SUCCESS in 3h 35m 36s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/5/20220907223746/NEC-FC : SUCCESS in 3h 35m 57s","accounts_in_message":[],"_revision_number":5},{"id":"83be3cf6c0c436263d0d63b39896d74ec22d1a15","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2022-09-08 19:46:51.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"91d70eb655898eb89b912ca4083a4453b04edc60","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-09 10:05:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3b7f8fb39848e9d88253f7bb11a5cf28538d9181","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-09 11:25:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a5a499f15f5041545b9e7167e75ce03a97a4c988","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-09-12 12:16:13.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a213e675f22d45ec8b9254f0d9228495 : TIMED_OUT in 3h 07m 27s","accounts_in_message":[],"_revision_number":5},{"id":"31bfb8dc69ded2b3cdd1f234837940ac51c246e7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-09-15 02:04:45.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/5/20220915050549/NEC-V-ISCSI : SUCCESS in 2h 13m 18s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/5/20220915050549/NEC-V-FC : SUCCESS in 5h 58m 51s","accounts_in_message":[],"_revision_number":5},{"id":"a1b2583dfab14e466d5b9f0788e5f7ff6ce02000","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-22 10:29:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"aaaa2203764c71e33a8a0b5ab182df12debe30e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 10:41:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e69af36247e74c44b48c9fc885e3ca19\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c0d3a5cb19a4d9abebcf4eb43350233 : FAILURE in 9m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1615ef1d71dc452bb93cba172607b433 : FAILURE in 9m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"65eb41d50c4072da4c4a4317be07a8c44cc3f8d0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-22 11:26:00.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/846173 : FAILURE in 23m 48s","accounts_in_message":[],"_revision_number":6},{"id":"895ebaf472586583f63df5d80db93f016a3a7940","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-22 11:30:09.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/73/846173/6/check/iscsi-driver-alletra-6k/637f39d/637f39d.html : FAILURE in 26m 12s","accounts_in_message":[],"_revision_number":6},{"id":"8bed3873a235794d4d7e9dcf177dd33820ce93b9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-22 11:31:38.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/846173 : FAILURE in 26m 06s","accounts_in_message":[],"_revision_number":6},{"id":"59647e3ba5d68a32e37dda7ab7136efbddf138fc","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-22 11:34:44.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/73/846173/6/ : FAILURE in 56m 40s","accounts_in_message":[],"_revision_number":6},{"id":"5287fad5dfb01aee6b74f0e7bffef1e468fd7d69","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-22 13:23:11.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/eb731727e97048d0b96e66f98a1295c0 : SUCCESS in 1h 12m 32s","accounts_in_message":[],"_revision_number":6},{"id":"f944a7d38df0fba5372e3b1a02821b3d992548e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 13:29:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f6a42229ab8485f84b436cfcf2af65f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee1e78543dda48a6af482476a78692f3 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf60ed6e36684e6d9885228b2388501e : FAILURE in 10m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/789ac6ff948a4dae8228435bb82331ea : FAILURE in 10m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b1110647c754e13a6fb5bd02487e7b7 : SUCCESS in 12m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/cd9d57b2aeb1403fa26963aa888f1756 : SUCCESS in 56m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c571b60d9edc4097a484905ac8d0af6f : SUCCESS in 58m 23s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a8f05df39543472da42b5f9c3670a7c5 : SUCCESS in 1h 31m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e530db9b0344cf68ff67140ef72963f : SUCCESS in 56m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/875df7e9e5aa4e2b987e0ad4b8582c04 : FAILURE in 12m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/79b7d2b2542f4237a141f3a45874c050 : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a035903c72d44fa98e6393749d876bb3 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5f50284b961b41a59715966be6102cfa : SUCCESS in 7m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5df8d292a9004cc08a65af29dd6a0a03 : SUCCESS in 6m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5ad75c410f29412c97c49304c0d6ada3 : SUCCESS in 50m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/30269c8d870d488bbc742449b2690af8 : FAILURE in 5m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e4a2cc59d6794e92b1459cc51d226880 : FAILURE in 1h 39m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b65a668faf924179b642207b85c940ff : FAILURE in 1h 43m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8f808546ece94b299209950d4a8dcd18 : SUCCESS in 1h 47m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/99d8416809ae4e8eab4497ec39444a33 : SUCCESS in 1h 18m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4a055af8eccf42278ae2e6bb83cf35b4 : SUCCESS in 1h 46m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e01750d9717d43cd81c154b41e464bf1 : SUCCESS in 33m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dbad71300d7e4a379bbdd3924a1536b1 : SUCCESS in 1h 11m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ae9d81174b6d410f9a7c40f35dd7cc9c : SUCCESS in 2h 33m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5386d97dbbc346d1ae4926bfc5f76587 : SUCCESS in 53m 03s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f307dff58504b978caa26fdf05c8a43 : SUCCESS in 1h 07m 28s","accounts_in_message":[],"_revision_number":6},{"id":"e4fbb33f309033239fe10a572bee17920b9a50f2","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-22 14:26:19.000000000","message":"Patch Set 6:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/73/846173/6/check/fujitsu-eternus-dx-iscsi/05ae7fc : SUCCESS in 1h 48m 06s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/73/846173/6/check/fujitsu-eternus-dx-fc/a94f31b : SUCCESS in 1h 45m 56s","accounts_in_message":[],"_revision_number":6},{"id":"4731b05907ce8520ed64aee82b9875169030602f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-22 16:32:11.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"043490a5073ee6e1275fb925a70a285ec404cc04","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-22 17:25:40.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/516ef054d66349b9b1fdcbfa39aeb902 : SUCCESS in 1h 10m 19s","accounts_in_message":[],"_revision_number":6},{"id":"81bb3d4543a0e904544a9e136e1187b8fcc922a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-22 18:52:58.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f4defb2e1c8d5ffbb559067c114a295e47c93974","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 19:06:35.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28f93da783154e0ea646bf10db220fa5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ef0fa2367c04bc1a59aa7e6e1007e19 : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2cb31e84599a4a14b961bc9c228f6c27 : SUCCESS in 8m 04s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"37d086653e894f1aedbfa397c2c7450fa805c389","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-22 19:19:04.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/846173 : FAILURE in 23m 35s","accounts_in_message":[],"_revision_number":7},{"id":"07b524cbc91c0318672decfa6288a481582251c8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-22 19:21:58.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"5901fc129fdd7d038df51d9898a1b02854519b14","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-22 19:22:16.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/846173 : FAILURE in 25m 56s","accounts_in_message":[],"_revision_number":7},{"id":"f6b0687dd1a23ce36b949e4be1d1239589863c49","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-22 19:22:32.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/73/846173/7/check/iscsi-driver-alletra-6k/f77ab37/f77ab37.html : FAILURE in 26m 09s","accounts_in_message":[],"_revision_number":7},{"id":"daaf3048f2bcb6a7bb07be918d439883e2507ab6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-22 20:10:26.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/afd8317bb14e4a738056c9aeef881a64 : SUCCESS in 1h 15m 07s","accounts_in_message":[],"_revision_number":7},{"id":"88f6109da0e4f1d7e093c6604f020c653c7abc72","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-22 20:24:01.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/73/846173/7/ : FAILURE in 58m 12s","accounts_in_message":[],"_revision_number":7},{"id":"c603463dee4f20bf7f0b55f0325e57a1d16bfe6f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-22 20:37:18.000000000","message":"Patch Set 7:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/73/846173/7/check/fujitsu-eternus-dx-iscsi/e6f757f : SUCCESS in 1h 39m 51s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/73/846173/7/check/fujitsu-eternus-dx-fc/6b5e0bc : SUCCESS in 1h 43m 23s","accounts_in_message":[],"_revision_number":7},{"id":"53f718f0f05e7ece09ab3343d1a74d7e49f8c897","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-22 21:10:11.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/73/846173/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a3bc27b/ : SUCCESS in 2h 05m 30s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/73/846173/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d2ec858/ : FAILURE in 2h 13m 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/73/846173/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/8b5d006/ : SUCCESS in 1h 34m 24s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/73/846173/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/0e1c2c9/ : SUCCESS in 1h 55m 39s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/73/846173/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/ed718cd/ : SUCCESS in 2h 10m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"26f4249fb992399f9ad1940b3311bf474d2f244d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 21:19:41.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f100966c052444fb8f1bf9d33ac6a6c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12f6d596da91442f9731b6fb2b17d0ef : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa146793ba994c0db04c24edea075068 : SUCCESS in 11m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7a5c8251c674a548c6912fd4aab02c8 : SUCCESS in 10m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1da07ff77c90446f80a6b1ba50ab2909 : SUCCESS in 14m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/f0c83295961c448e8fee43841981b5a0 : FAILURE in 1h 03m 05s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0ebda45839fc446e859fdfe7c224f67b : SUCCESS in 1h 07m 36s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/239e5e8de8b54a28a6061f3c3df1ce65 : SUCCESS in 1h 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9b5263b1e46423682fe7308a65722ae : SUCCESS in 55m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/daf8e8b959e145af9fbd188530a9d528 : SUCCESS in 14m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4dba816450e449efbfe9fe3a3c069a1e : SUCCESS in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e3474585431c442ca8fee2d29c7b5949 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c624ecbe70f841f8a138dc7c2b6e05bb : SUCCESS in 6m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/730fe7160dbc4e968f16c660a7e5e162 : SUCCESS in 6m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5f77e8b3da0d4d4c802591ba0bcf370f : SUCCESS in 35m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1ea06289df424fe48c3affceee306740 : FAILURE in 7m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/98583470c5a04c21be48083f45308979 : SUCCESS in 1h 29m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3374364f5a5347fab556a72a88b5e48a : FAILURE in 1h 23m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b9f370e0fbd34639a6bf460a7a48d04c : SUCCESS in 1h 34m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/07c17859a6b0402b847cacb946015dce : SUCCESS in 58m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/32e3f3c0b4614e5f8da36459119370cd : SUCCESS in 1h 29m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/45e1d0db296342a5bf651e9020209aa5 : SUCCESS in 1h 12m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4f0e74822c1c4fd3a501dfca5623d3a5 : SUCCESS in 1h 23m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/66b8782dd05a4ad8bc408ee66435a2a6 : FAILURE in 1h 47m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdb93ec606ed4eb8b0a3beef53f40bc4 : SUCCESS in 51m 12s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1a37a775d984509ae3b62c7fb4a7561 : SUCCESS in 1h 16m 43s","accounts_in_message":[],"_revision_number":7},{"id":"4d5ce24f0d20f1fcfe7170bdf6d75c0f4454ede0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-22 21:29:33.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-73-846173-7/ns5_nfs : FAILURE in 1h 48m 11s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-73-846173-7/ns5_iscsi : FAILURE in 1h 48m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"1f7e1fc3f0495375d33f860744287a95e0b24a1f","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-22 23:31:20.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/e4a5d456e8e94e6db644a9f5d48d125f : SUCCESS in 1h 14m 38s","accounts_in_message":[],"_revision_number":7},{"id":"180ad65ae812d0dd213d39841cfb51d1a080050f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-23 00:49:34.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-73-846173-6/ns5_nfs : FAILURE in 13h 34m 12s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-73-846173-6/ns5_iscsi : FAILURE in 12h 51m 52s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"6a1cf9df0ba894c830cdc0904aa546c9995e2b74","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-23 07:27:39.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/73/846173/7/check/DellEMC_Unity_FC/d0f4e37/DellEMC_Unity_FC/2352 : FAILURE in 1h 40m 15s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/73/846173/7/check/DellEMC_Unity_iSCSI/26f1ed7/DellEMC_Unity_iSCSI/2205 : FAILURE in 2h 16m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"03bc22203c0a7e8b1b813c0fed921cb749cd6e23","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-01 08:05:32.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"ecaff25b0475c6fea35775e236af5e430f0bd980","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-03-02 21:55:16.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/7/20230303032248/NEC-ISCSI : SUCCESS in 3h 31m 39s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/7/20230303032248/NEC-FC : SUCCESS in 3h 32m 25s","accounts_in_message":[],"_revision_number":7},{"id":"2d39bf23d1f9c1356d40957c9bdbd2b4c5657c4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-03-30 16:15:04.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"823dc6beffedee1470e16374fa6a21a1d5131986","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 16:45:04.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/163871be232e4d0eb38abcff309a6580\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/695afde8b57045158fcbeea8054af60a : FAILURE in 27m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c534f566b9e04690a785a74aad3bdd13 : SUCCESS in 9m 15s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"603f0fa8cc8b4bc133ea791f584e856166da10c3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-30 17:08:32.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder finger://openstack-ci-fc/f600a5616dd74a44968e66e157779d21 : RETRY_LIMIT in 6m 31s","accounts_in_message":[],"_revision_number":8},{"id":"b60144779b5355fb8184eb677cb80089b7aced75","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-30 18:49:26.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/54b7a73722384bfabc80926092a18f16 : SUCCESS in 1h 14m 56s","accounts_in_message":[],"_revision_number":8},{"id":"485860b05b2ac4b65c1e9975a2e52c6e90c8cf7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 19:03:24.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/081c73ffd3da4cdc865d16850b7c55ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c50e93d0ebcd4a348c5387af60dfcfa1 : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2250a49e12bd4b0d8f9034cc972ce876 : SUCCESS in 11m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17e6b7f0192f43ddb4c113815e6894d7 : SUCCESS in 12m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc5039c8f37a414d8f13ff25fbee5773 : SUCCESS in 14m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/385664f75a7c4bc188e5125dd124136b : SUCCESS in 1h 15m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6655a289aed94be5942846df49967a33 : SUCCESS in 1h 32m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/146f26d2fe85404a8913d841a3cd5bf9 : SUCCESS in 55m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/75b9a9aebb59494883d29fb2ad4ecef9 : SUCCESS in 17m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e0cd5d1788524269b08a41edb69d3178 : SUCCESS in 4m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/22ff27d4bb8c444fafe73b3f11255a04 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/36389a59917f46cd92ae4ca299b257d0 : SUCCESS in 5m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/633bb2f3b5064fb68b22c7e139fe9af0 : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/809cacf5b037470caed64dec95bdd77a : SUCCESS in 51m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a4d7673c842a4e9099980efe47e627be : FAILURE in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e7031b3806464510b9eeffbe3832c3f0 : SUCCESS in 1h 06m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5b5451f825564b2ba9ea6fbc4e5855b7 : FAILURE in 1h 32m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/de6def17d3bf4b19a7da17377ce08605 : SUCCESS in 1h 42m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/66e9a5cf815542439fcaafb6ce9052d7 : SUCCESS in 1h 16m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/46e931b531b44c06a995ed25a19f1d95 : SUCCESS in 1h 46m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/28929a24db5b40b3b30a9ad7539754f2 : SUCCESS in 42m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/25e79cda148449cfa1ff42c57299dacd : SUCCESS in 1h 05m 16s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6446e85c75914b139eee6d1ea9416afd : SUCCESS in 2h 31m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b4c56fd36a74c41be82ae768c8237dd : SUCCESS in 45m 42s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a046278dc010494486e64e9e83aac087 : SUCCESS in 1h 17m 00s","accounts_in_message":[],"_revision_number":8},{"id":"886f25cb36983f921f31303b55f3fbe7a66202c4","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-03-30 19:04:42.000000000","message":"Patch Set 8:\n\nBuild failed.","accounts_in_message":[],"_revision_number":8},{"id":"01cfa65728b20fe358388e76713e222e7f9cdfaf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-03-30 19:11:26.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_03_2023_19_03_33\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"95cf3f08119c2fbf9db64b19931ef321471604f0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-30 20:47:36.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/846173 : FAILURE in 1h 28m 06s","accounts_in_message":[],"_revision_number":8},{"id":"b7062deb70a76b7867dd2a86ab660204e1ce7ad5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-03-30 20:52:03.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp cinder-tempest-plugin-ontap-fcp : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/846173/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/bf63038/ : SUCCESS in 45m 30s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/846173/8/upstream-check/cinder-tempest-plugin-ontap-nfs/f200172/ : SUCCESS in 44m 20s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/73/846173/8/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/87950c5/ : SUCCESS in 41m 22s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"b60ce3900fd9d2809da38f9a95e834e03e18f567","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-30 20:54:30.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/73/846173/8/check/iscsi-driver-alletra-6k/51fbcc0/51fbcc0.html : FAILURE in 1h 15m 16s","accounts_in_message":[],"_revision_number":8},{"id":"1459d30b24b8f7fbee02012d9fa0f4f067d7896d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-30 21:11:08.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-73-846173-8/ns5_nfs : SUCCESS in 3h 28m 47s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-73-846173-8/ns5_iscsi : SUCCESS in 3h 27m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"f5a6d561b8cac66fa03014b502ee19485e46cb43","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-30 23:56:48.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/73/846173/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/af1e439/ : SUCCESS in 1h 54m 56s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/73/846173/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/336b2e3/ : FAILURE in 2h 25m 00s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/73/846173/8/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/473b935/ : SUCCESS in 1h 43m 47s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/73/846173/8/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/1afb533/ : FAILURE in 29m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"45e653d8fd6ddbfb03677fa8f81ceb55915da96e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-03-31 00:00:10.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/73/846173/8/ : POST_FAILURE in 1h 07m 35s","accounts_in_message":[],"_revision_number":8},{"id":"01887d0bff63f44a9c28e4229c16acd1e6f9d6a5","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-03-31 03:11:45.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-73-846173-8/FC/ : SUCCESS in 3h 13s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-73-846173-8/iSCSI/ : SUCCESS in 1h 51m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"ef8322ef42f15c2a19799357242397c39edecbf7","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-31 06:32:01.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/4ce7d00c6ca74b85958aaa78104bf647 : SUCCESS in 1h 20m 07s","accounts_in_message":[],"_revision_number":8},{"id":"cdfdb0966ab955c8940abe94f9b6b85a9730b7a9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-31 06:48:27.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/73/846173/8/check/DellEMC_Unity_FC/d9cd783/DellEMC_Unity_FC/2611 : SUCCESS in 1h 40m 26s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/73/846173/8/check/DellEMC_Unity_iSCSI/5a956fb/DellEMC_Unity_iSCSI/2458 : SUCCESS in 1h 52m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"49f318a60df6b5761567d1a498ddd1f00ad22666","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-03-31 07:22:15.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/73/846173/8/check/DellEMC_PowerFlex/88bc7ca/DellEMC_PowerFlex/1806 : SUCCESS in 1h 34m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"f814bd670aa50ee34aa2172271dd93796c12f230","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-31 17:23:40.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/73/846173/8/check/tempest-dsvm-full-focal-py3/d8b5c70/ : SUCCESS in 2h 18m 31s\n- devstack-plugin-ceph-dsvm-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/73/846173/8/check/devstack-plugin-ceph-dsvm-focal-py3/61a4c27/ : SUCCESS in 2h 19m 33s","accounts_in_message":[],"_revision_number":8},{"id":"a2be009ae9b9c3adac14368e63517a412adc0e05","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-04-01 06:54:16.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/846173-8/HPMSA-iSCSI/ : SUCCESS in 1:34:47\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/846173-8/HPMSA-FC/ : SUCCESS in 1:17:34","accounts_in_message":[],"_revision_number":8},{"id":"1b5a0febf6be5c9d66c92700b5d3e847f40382e1","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-04-01 10:53:35.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/846173-8/PVME-iSCSI/ : SUCCESS in 1:44:22\n- PVME-FC https://os-logs.seagate.com/dell-ci/846173-8/PVME-FC/ : SUCCESS in 1:27:48","accounts_in_message":[],"_revision_number":8},{"id":"8f729b0322173a0c26c33b28b5a5ad3f30145122","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-04-02 00:27:39.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/8/20230402063444/NEC-V-ISCSI : SUCCESS in 2h 25m 13s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/8/20230402063444/NEC-V-FC : SUCCESS in 2h 52m 51s","accounts_in_message":[],"_revision_number":8},{"id":"0484c2e750dce694f5655e35bbec873ee6491760","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-04-06 08:28:40.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/8/20230406134902/NEC-ISCSI : SUCCESS in 3h 26m 30s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/73/846173/8/20230406134902/NEC-FC : SUCCESS in 3h 39m 34s","accounts_in_message":[],"_revision_number":8},{"id":"e5db414c4434b92a47d73bcc53b3fc8d833441ec","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-18 17:29:53.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/73/846173/8/check/iscsi-driver-nimble/163dfed/163dfed.html : FAILURE in 1h 03m 01s","accounts_in_message":[],"_revision_number":8},{"id":"8fd38f58bcf6288a3b9b86fe9c29b99e2f79682a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 06:09:09.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"597e36d1f173504eb6b613f1d98a0f08cc86bfa0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-22 06:09:23.000000000","message":"Patch Set 9:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/982/","accounts_in_message":[],"_revision_number":9},{"id":"0936913646add9f7e7611ebd914dda3275ac5351","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-22 06:10:14.000000000","message":"Patch Set 9:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/846173/9/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/846173/9/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"98541eb6a6ebdb3053f89ab6892e2e6e18ff3688","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 06:17:20.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7e893ff321a55204eefba010e2552c37c2177354","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-22 06:17:30.000000000","message":"Patch Set 10:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/983/","accounts_in_message":[],"_revision_number":10},{"id":"f7dd1acf8fd20a0cffc8613ceb3d60b18babb721","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-11-22 06:18:25.000000000","message":"Patch Set 10:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/846173/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/846173/10/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"0826301be03189fa782bd5df271bf04a9039af86","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-11-22 07:35:26.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/846173 : FAILURE in 34m 35s","accounts_in_message":[],"_revision_number":10},{"id":"4a7d448b74c69a2c7465153b56c363d39eee129f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-11-22 07:48:18.000000000","message":"Patch Set 10:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d53614401b534d599f02e5177883de42\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/81a3167971bb416babb471a3e3311705 : SUCCESS in 41m 23s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ca5cfae699d74c2886f48cedcd176b3b : SUCCESS in 42m 17s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6244078bfab74779a1aeb754d7f234f1 : FAILURE in 1h 30m 14s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/88140d418009431cb4ba24830c26c560 : SUCCESS in 39m 35s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bfaa44158fde4d6abc6da7a622bd74f4 : RETRY_LIMIT in 30s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b605f21c5c0245339728787386b1020c : RETRY_LIMIT in 29s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6d49b8dfab834bb59a755e9bef807b5a : RETRY_LIMIT in 31s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1fd5ce0015ca4e949ce32080be2d1022 : RETRY_LIMIT in 31s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.1}} source: openstack/devstack/.zuul.yaml@stable/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pur... (truncated)","accounts_in_message":[],"_revision_number":10},{"id":"9ac314097c10475a9027b6419c96672c88e0a2e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 08:01:41.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17d4a40e11bf43e59958f66974d53d73\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/754b4e095f46403eb9cd113c2fb450f6 : FAILURE in 30m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/15d0e83ebbfc44e08281c084baa2f0a3 : FAILURE in 28m 25s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0f33fb43116fd15a96efe2def216ce632ba173d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 08:35:52.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cceaf7ec5744d19a380c19d9b294363\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/923fca00655c4f5eba8dd3a9489634a8 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43b2a12ceb144e77aa811a55b5ba4303 : SUCCESS in 11m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a6d3e9e09044b339a575b1e17c1cf98 : SUCCESS in 13m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aae38812ef54680b94c6eec2890d89a : SUCCESS in 11m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/21235c1386b34bc287de0bbdfa25a642 : SUCCESS in 57m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5370c108601242008eb1e896043e7d6e : SUCCESS in 45m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bf69e822885f4e1a9d3f73c3d0e8ab2f : SUCCESS in 1h 21m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d754b3c014746d98d337ef72b8dac8d : SUCCESS in 50m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/90d96901c3604e9e89a162c6a9e1a93d : SUCCESS in 16m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b7ff245875ef4a61a3b3f26efec96580 : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6b420383967a4a2eb9ee0f98748cbb3b : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ad6b7d2ede7a4074b9a8618825d118de : SUCCESS in 5m 42s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/519b8a555cb64f73943fe62aee0c9d2b : SUCCESS in 7m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/60ebe79732f94cb18250e64d96067dd2 : FAILURE in 20m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fc4f927202104b938736c2605159e4c0 : SUCCESS in 6m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3e8caf6f97c7469887253183776c8bbf : SUCCESS in 1h 52m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2a9d24b5ad9e413697f276b863a232b2 : SUCCESS in 2h 09m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eeeaff14dbb24240bc66c0eb87d799ae : SUCCESS in 2h 17m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2c11081c79694a95bd615b45803663ca : SUCCESS in 2h 01m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/59595ad1a3a9436086def02266e22d30 : SUCCESS in 27m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0e13dabb38284281a56de08c1c70e9c0 : SUCCESS in 1h 04m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/de60c471d66e4278b6fa6719dccb9615 : SUCCESS in 1h 13m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/412f4372ea814203a7eabac1d3bb06cb : SUCCESS in 28m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/903c3d5aa52d4645acd035e3e37b8be4 : SUCCESS in 1h 18m 22s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f6ce452823f744beb492d72a144c59e8 : SUCCESS in 58m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fda3808d01cd4c4bbb8ece32e63c3936 : SUCCESS in 58m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/697537cdd09f4d0f94278d1f3b6b7c05 : SUCCESS in 33m 31s","accounts_in_message":[],"_revision_number":10},{"id":"e45ee90eca07dce378109aa22c4f93c6b30d7328","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-11-22 08:41:17.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/60360fdffb2047fcaab8752969b58ada : SUCCESS in 1h 13m 09s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9766d95918204406908a04466951321c : FAILURE in 1h 19m 00s","accounts_in_message":[],"_revision_number":10},{"id":"28c65fd1826d3f554e4743c688d2e37b7c0dff7d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-11-22 08:52:44.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/22955bff47f4459489119dc979defc29 : RETRY_LIMIT in 5m 42s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"c57cf15a99bd0514c725383fae8d56802ab0cb8f","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-11-22 08:54:17.000000000","message":"Patch Set 10:\n\nBuild failed.","accounts_in_message":[],"_revision_number":10},{"id":"f912f4f830128018902addb2f7535f3e9496025e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-11-23 04:31:51.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/73/846173/10/check/DellEMC_PowerFlex_3.6/3df35aa/DellEMC_PowerFlex_3.6/2484 : FAILURE in 1s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/73/846173/10/check/DellEMC_PowerFlex_4.x/148aa46/DellEMC_PowerFlex_4.x/1322 : FAILURE in 1h 48m 54s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/73/846173/10/check/DellEMC_PowerFlex_4.x_NVMeTCP/4437b16/DellEMC_PowerFlex_4.x_NVMeTCP/1208 : FAILURE in 4h 32m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"7fceaa85cd28458d18201dd2cb527c5c14813114","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-11-24 00:17:45.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/73/846173/10/20241124050826/NEC-V-ISCSI : SUCCESS in 3h 55m 17s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/73/846173/10/20241124050826/NEC-V-FC : SUCCESS in 4h 09m 16s","accounts_in_message":[],"_revision_number":10},{"id":"b65a339f300cdd175c90d7634eac9120dc3c49c3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-11-24 17:11:53.000000000","message":"Patch Set 10:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/73/846173/10/20241124175939/NEC-FC : SUCCESS in 6h 04m 33s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/73/846173/10/20241124175939/NEC-ISCSI : FAILURE in 8h 12m 11s","accounts_in_message":[],"_revision_number":10},{"id":"b7f5b3ce94f5fde3069a44710c31957d8ebedd36","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-13 22:10:49.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_01_2025_22_05_11\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"94bb2e585b458fc2387aa37ea230949b66f1a8fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 13:05:07.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"ab261265568ab765647cb2abff234b4dca85a939","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 13:05:19.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"0c4abd35c10f3a9e6ab619cb2f342df15e50c016","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-10-24 13:25:11.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":11},{"id":"3cf036ae091e43b401a2090047db5c39ede7f6db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 15:32:59.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a89c627713314d6f8e1664e4ec90e931\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e61ddf03e674059b3eb6ca027e1de94 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ede07b360a814f359611810431d3b765 : SUCCESS in 11m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/209601a41b004a4c8d416e66606962be : SUCCESS in 14m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8e043f644fd481d882b5e8d2d63c732 : SUCCESS in 14m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/63f4522791414ecd93473dd9e947b35d : SUCCESS in 59m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/10ff3d2cb14040a7a8481e5360e693dc : SUCCESS in 1h 00m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cfe14889e1f54f38a708e09557216fdb : SUCCESS in 1h 20m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42150525b2e64e04959279635fdb953d : SUCCESS in 58m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fcc728801dd64b4e943dc8cea5860abd : SUCCESS in 32m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c151f664309b4fe4b391a20a4540cd28 : SUCCESS in 6m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cba701da31ef4b12ac3b4c8ed1ea3d6d : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b347594904dd4c25b5ff033b93ee6315 : SUCCESS in 6m 31s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1715e1e102e74a13af4e98ed58daa49b : SUCCESS in 5m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f080c35da2f6442dbf26a8a0b63ac3f9 : FAILURE in 29m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6d0aee28755e4cb3b7ab967bbd8b2491 : SUCCESS in 6m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c4cbf35455d941de86a91613034e8f02 : SUCCESS in 2h 03m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/154bdccbac894aca8552f949bca71d7f : FAILURE in 1h 30m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ae0783beb82c4ca1b44d7dd804b692ac : SUCCESS in 1h 56m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/85d2858603014a6897d288a0933dbcdf : SUCCESS in 1h 38m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f8d44b4265fc4207bfc90869fd50bc65 : SUCCESS in 39m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/00200c31895b4467bc5228127ea32a8b : SUCCESS in 1h 13m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f0105538ce6c4783aae3b39a212156d3 : SUCCESS in 1h 33m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/95a81b1294854e30a08eb3aae6c547c4 : SUCCESS in 48m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d7a454a56a34452a8e66d753bc7875ac : SUCCESS in 1h 05m 08s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/f39a2e318cba4b65af7b55f9d782f866 : SUCCESS in 2h 09m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bbc8af97ca5d45e5a6df7cf83cc3595f : SUCCESS in 1h 47m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c909da4d42e4f6b9b6fbd3cc8b20d25 : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":11},{"id":"9199021ac50cd2ae371ac1e5bd954fbe73361340","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-24 15:43:37.000000000","message":"Patch Set 11:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_10_2025_15_37_08\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":11},{"id":"65a9249ef650428b76c5455530e51449122599aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 16:42:21.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d816d66c999143c1b916d889474b21e0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/40de40300c4945318bbb4af236998a67 : SUCCESS in 20m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c7aeb0f4e5fb4c988c3b94c744b1a31b : SUCCESS in 14m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ea277c601dbf401d9fb3d45f8d0d6e0f : SUCCESS in 29m 03s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c3a2233eaac2ee7261739060e440ed235913edf9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-10-24 17:40:02.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci (This CI is currently under construction) : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4 .","accounts_in_message":[],"_revision_number":11},{"id":"993761b3836901b5fe4d6353c500403290f64276","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-10-25 08:37:25.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/43233b8dc1644c31bb3ee02c88a0cc4c : SUCCESS in 42m 51s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c5b31a5657a745d6ad312a1c9a865250 : FAILURE in 52m 54s","accounts_in_message":[],"_revision_number":11},{"id":"9a87ceeffd57e2437481b843a788c08e51103bbe","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-10-25 12:08:06.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/846173 : FAILURE in 35m 50s","accounts_in_message":[],"_revision_number":11},{"id":"8a06ed99a1f915f11e5c64cde568b67ccc22d873","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-26 00:39:17.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2169/.","accounts_in_message":[],"_revision_number":11},{"id":"4b69f32c213436afce47590155a79873a2f41324","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-10-26 00:48:09.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2283/.","accounts_in_message":[],"_revision_number":11},{"id":"6554a2af71249d7a242a868039e50161db6eabf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-12 13:42:46.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"31f770809a2042815616c4faabc257abf2cb4b5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 14:02:50.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fab3d6b6ef4405fb52ee0a206fd9496\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7aa2d20e42b245b9805ae5065165603d : FAILURE in 5m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/17ce2431b470456d9ba3319a73897952 : FAILURE in 7m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fcc239fdde304986b767979de0e41f7f : FAILURE in 8m 15s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b335e3013c5274d4e2e175bfa2469526b5e08107","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-12 14:45:02.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4 .","accounts_in_message":[],"_revision_number":12},{"id":"b413a825f14e06af02c6594efa230808954dc4de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-12 15:42:04.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"40a22b469e065d84e0b1301e0dec1fb868792b86","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-12 16:39:21.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4 .","accounts_in_message":[],"_revision_number":13},{"id":"33201d1344a3117b827cce18f32ed31caefeea08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 17:32:53.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8009d9a5850c4a8fa224d04a7dc4a3c4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/949a13459b3f479780f44751840140c2 : FAILURE in 5m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f88a666133df45f091d1201149914727 : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/62e5c951de0543a8808ea67accb075b0 : TIMED_OUT in 1h 01m 24s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"e3626f66bc51994bd522281329a18447b0a9a970","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-05-12 17:53:08.000000000","message":"Patch Set 13:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/846173 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"9d91cee30b982c3476763ca4e722108c565f9f09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 18:23:29.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90bb18c8754c4a6aa5a26e7e7071beda\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/417cb0e969f54b9595266bf475af05d9 : SUCCESS in 8m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/375ffa0d004a4df0a0dc61014f306c0a : SUCCESS in 10m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ffed18e2af64344b45df696bced6e37 : SUCCESS in 10m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/952c4368e6c4471ea2ea5099189552f8 : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/428301255a0345369abbd7e5fa383359 : SUCCESS in 15m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/25d880bb042d4329a94610e72a9fe4ba : SUCCESS in 1h 07m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bc417c71fe394d4ab39ad15f0f3faaa6 : SUCCESS in 33m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/29ce3b4e71d540c59dff1f0d9baf6a0d : SUCCESS in 1h 56m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ed7705f184942ff8025ee664c3615d2 : SUCCESS in 53m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78b6eca7b06f4981bb3dfaf1d72d02f0 : SUCCESS in 6m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/64bbe4d350664d029879e9302847f3e7 : SUCCESS in 24m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4fc10b3b5e2b448cb0b7a2735a798917 : SUCCESS in 5m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/265004a869be4f6b84f24e2159e872ac : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b75380e6865f4c048771dd38aaac0355 : SUCCESS in 6m 12s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/444763fea2b443058762422b08592944 : SUCCESS in 6m 45s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/35b5e96d7f374ef58776565bab070f40 : SUCCESS in 9m 07s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/25301aab86fe4849b94066712a6e54b7 : SUCCESS in 41m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e19d7cd829ac45b689d885d3d7cb7861 : FAILURE in 6m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ff8f6c504e2c41e8ba4972f90a771f17 : SUCCESS in 1h 58m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0e0c3f8d81de45eea8a9e998235366a5 : FAILURE in 2h 23m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3513d264e4d7498ea0b25789b2f27981 : SUCCESS in 2h 10m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4d06bd8315ff495aac8f9150cf7783ff : SUCCESS in 1h 58m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/65f18efcfd5e49b695429517c61660c9 : SUCCESS in 46m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f95a2514763b4c9faac93207650448c0 : SUCCESS in 1h 08m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e60407294ad64a7b92388ccb0aa47b32 : SUCCESS in 1h 31m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0aeb9305d3294934b90b9b16f00509af : SUCCESS in 46m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/64f937007853486c8d4e2e70ba91847f : SUCCESS in 1h 23m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/29ec8106daf142dab28b63d288482ff3 : SUCCESS in 1h 54m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba4be892c7924fb39e86fe44627b8041 : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":13},{"id":"4c3f9b7869de46a8da02edaa3020134e95adc744","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-12 18:33:42.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/12_05_2026_18_27_10\nLeave a comment with","accounts_in_message":[],"_revision_number":13},{"id":"4793bf66993864a4ad49155c08da3ec5ce4e2491","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-12 18:48:37.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-fc-driver build/4359e82f1ba24725b733af8ee0c7547c : FAILURE in 38m 29s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/846173/etc/tempest_log.log","accounts_in_message":[],"_revision_number":13},{"id":"1356f4307520759daaeeb4269b126d9dacf818f4","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-12 20:19:51.000000000","message":"Patch Set 13:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/846173 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"f94e987b9f565b977051f0ac0ef262da6544666d","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-05-12 20:25:45.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-73-846173-13-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-73-846173-13-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"d90313cb34d5dfd53031ca301437fe9dafdb4ee5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-05-14 04:10:28.000000000","message":"Patch Set 13: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"3315c8375b23248ab6a1d93e53253b2f167088c7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-16 20:11:29.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/c2a4d7500a7646fea00e302e06237b29 : FAILURE in 31m 14s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/846173/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/846173/etc/tempest_log.log","accounts_in_message":[],"_revision_number":12},{"id":"78f5d01d6a464d4769661d906f119828ee7caa8f","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-06-04 15:28:01.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"ea45f2ea782e24f7f0d5f7fac2d8e411aef5ab90","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2026-06-04 15:28:19.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/837543 is needed.","accounts_in_message":[],"_revision_number":13},{"id":"acb3637fa0753094c1a016dced4e837a017f4392","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 15:28:46.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"a8f0772f02e3f1e5a4a7e9c4f2c0d2f3bd388c59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 18:20:32.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a7f4ea07f614c3c94f6fbc364aee54d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/739644f211c84e7aa959856bbb01dc73 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d523590e9c4c4c6a93c38f07a04e610d : SUCCESS in 10m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/503e8c47d8564755a15941c7638b9ad2 : SUCCESS in 9m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4c986ac854c48f2999b64659a387029 : SUCCESS in 14m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/23018d9ab1ef4cd68ff84fb5261e1ae0 : SUCCESS in 1h 06m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9ac476d54a26477383e24321772091f1 : SUCCESS in 1h 03m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e1661e4a8d544358cbb0b98e09414c4 : SUCCESS in 1h 36m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b02e77682cac43cd989c34105f049054 : SUCCESS in 1h 03m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7e124ad302d49cc8cdef45a3120c5df : SUCCESS in 9m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/75eef387ac3b4118bbc9ae0dc075e272 : SUCCESS in 1h 12m 08s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ac96f81d4c174222bf2888a6bdf8728f : SUCCESS in 1h 31m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d5e59b2531f43148c76e404f33fb57f : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":13},{"id":"ada4380ac48e12960f24885ff39fdd412d345052","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 18:20:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"b6f5c8ffcd734d6b02796a001830449f3d72067c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-04 18:20:44.000000000","message":"Patch Set 13:\n\nBuild Failed \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/2092/ : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"486188e9a55c04a3f27bbac3787cb8fce98b68fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 18:23:42.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc04bcc9cede40dc9c431334bf94a4e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30bc6860ea5546309be3937af622d057 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":13},{"id":"9a86fa0a80ed80d8cdb2b842dba415a1bf2afa7e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:00:52.000000000","message":"Patch Set 12:\n\nBuild Failed","accounts_in_message":[],"_revision_number":12},{"id":"c323ea7be25631c8d66702189296d3ba300f53be","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:03:52.000000000","message":"Patch Set 13:\n\nBuild Failed","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"43564068e28905683750522d9fed4f14bf5fa580","revisions":{"6d2429507496f708d2c3412e9f10dba4922c1c70":{"kind":"REWORK","_number":1,"created":"2022-06-16 13:54:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/846173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/846173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/846173/1"}}},"commit":{"parents":[{"commit":"09a8535de57cafc85a9f41090def86d358c7799e","subject":"db: Move helpers to cinder.db.sqlalchemy.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/09a8535de57cafc85a9f41090def86d358c7799e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-15 11:42:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-16 13:53:13.000000000","tz":60},"subject":"db: Use cinder.db.api directly","message":"db: Use cinder.db.api directly\n\ncinder/db/__init__.py was importing * from cinder.db.api. This meant that\nany time any code anywhere within the cinder.db package was imported\nthen cinder.db.api was too, leading to a cascade of imports that may\nnot have been desired. Also, in general, code in __init__.py is a pain.\n\nTherefore, this change adjusts code that so that either:\n\n* cinder.db.api is used directly\n* cinder.db.api is imported as \u0027db\u0027\n\nIn either case, the functionality remains the same.\n\nThe primary goal of this change was to make it possible to import the model\nfiles without having to import the DB API. This looks like a very large\nchange, but it is essentially adjusting package names.\n\nNote that nova did this some time back in change\nIc1fd7c87ceda05eeb96735da2a415ef37060bb1a.\n\nChange-Id: Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d2429507496f708d2c3412e9f10dba4922c1c70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6d2429507496f708d2c3412e9f10dba4922c1c70"}]},"branch":"refs/heads/master"},"977cadca87d1c4ee3434eb1cc334ce859a51869d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-06-17 10:50:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/846173/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/846173/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/846173/2"}}},"commit":{"parents":[{"commit":"42d7feaac6c9795fbef69fce78bbf71780472f65","subject":"db: Move helpers to cinder.db.sqlalchemy.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42d7feaac6c9795fbef69fce78bbf71780472f65"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-15 11:42:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-17 10:50:20.000000000","tz":0},"subject":"db: Use cinder.db.api directly","message":"db: Use cinder.db.api directly\n\ncinder/db/__init__.py was importing * from cinder.db.api. This meant that\nany time any code anywhere within the cinder.db package was imported\nthen cinder.db.api was too, leading to a cascade of imports that may\nnot have been desired. Also, in general, code in __init__.py is a pain.\n\nTherefore, this change adjusts code that so that either:\n\n* cinder.db.api is used directly\n* cinder.db.api is imported as \u0027db\u0027\n\nIn either case, the functionality remains the same.\n\nThe primary goal of this change was to make it possible to import the model\nfiles without having to import the DB API. This looks like a very large\nchange, but it is essentially adjusting package names.\n\nNote that nova did this some time back in change\nIc1fd7c87ceda05eeb96735da2a415ef37060bb1a.\n\nChange-Id: Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/977cadca87d1c4ee3434eb1cc334ce859a51869d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/977cadca87d1c4ee3434eb1cc334ce859a51869d"}]},"branch":"refs/heads/master","description":"Rebase"},"39ae3d39f7977fd98d11cd47cacc5eae92f05a8e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-07-18 15:45:39.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/846173/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/846173/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/846173/3"}}},"commit":{"parents":[{"commit":"411bb9e7c4214b6d9c1a7aeeee52dfa400e126a0","subject":"db: Move helpers to cinder.db.sqlalchemy.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/411bb9e7c4214b6d9c1a7aeeee52dfa400e126a0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-15 11:42:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-18 15:45:01.000000000","tz":60},"subject":"db: Use cinder.db.api directly","message":"db: Use cinder.db.api directly\n\ncinder/db/__init__.py was importing * from cinder.db.api. This meant that\nany time any code anywhere within the cinder.db package was imported\nthen cinder.db.api was too, leading to a cascade of imports that may\nnot have been desired. Also, in general, code in __init__.py is a pain.\n\nTherefore, this change adjusts code that so that either:\n\n* cinder.db.api is used directly\n* cinder.db.api is imported as \u0027db\u0027\n\nIn either case, the functionality remains the same.\n\nThe primary goal of this change was to make it possible to import the model\nfiles without having to import the DB API. This looks like a very large\nchange, but it is essentially adjusting package names.\n\nNote that nova did this some time back in change\nIc1fd7c87ceda05eeb96735da2a415ef37060bb1a.\n\nChange-Id: Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39ae3d39f7977fd98d11cd47cacc5eae92f05a8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/39ae3d39f7977fd98d11cd47cacc5eae92f05a8e"}]},"branch":"refs/heads/master"},"9d649ef53f2b691fb8cd53573592e102f313fd40":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-07-28 16:44:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/846173/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/846173/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/846173/4"}}},"commit":{"parents":[{"commit":"42f2d032e2267c1d609238a959d9cdaababba31f","subject":"db: Move helpers to cinder.db.sqlalchemy.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42f2d032e2267c1d609238a959d9cdaababba31f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-15 11:42:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-28 16:29:56.000000000","tz":60},"subject":"db: Use cinder.db.api directly","message":"db: Use cinder.db.api directly\n\ncinder/db/__init__.py was importing * from cinder.db.api. This meant that\nany time any code anywhere within the cinder.db package was imported\nthen cinder.db.api was too, leading to a cascade of imports that may\nnot have been desired. Also, in general, code in __init__.py is a pain.\n\nTherefore, this change adjusts code that so that either:\n\n* cinder.db.api is used directly\n* cinder.db.api is imported as \u0027db\u0027\n\nIn either case, the functionality remains the same.\n\nThe primary goal of this change was to make it possible to import the model\nfiles without having to import the DB API. This looks like a very large\nchange, but it is essentially adjusting package names.\n\nNote that nova did this some time back in change\nIc1fd7c87ceda05eeb96735da2a415ef37060bb1a.\n\nChange-Id: Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d649ef53f2b691fb8cd53573592e102f313fd40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d649ef53f2b691fb8cd53573592e102f313fd40"}]},"branch":"refs/heads/master"},"25fc8fc3f30b483d4f8a63118c0e532e5fb8b2d6":{"kind":"REWORK","_number":5,"created":"2022-08-31 08:56:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/846173/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/846173/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/846173/5"}}},"commit":{"parents":[{"commit":"e90ce7134e23bee181d7824400f569a8cb72e82b","subject":"db: Move helpers to cinder.db.sqlalchemy.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e90ce7134e23bee181d7824400f569a8cb72e82b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-15 11:42:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-08-31 08:53:50.000000000","tz":60},"subject":"db: Use cinder.db.api directly","message":"db: Use cinder.db.api directly\n\ncinder/db/__init__.py was importing * from cinder.db.api. This meant that\nany time any code anywhere within the cinder.db package was imported\nthen cinder.db.api was too, leading to a cascade of imports that may\nnot have been desired. Also, in general, code in __init__.py is a pain.\n\nTherefore, this change adjusts code that so that either:\n\n* cinder.db.api is used directly\n* cinder.db.api is imported as \u0027db\u0027\n\nIn either case, the functionality remains the same.\n\nThe primary goal of this change was to make it possible to import the model\nfiles without having to import the DB API. This looks like a very large\nchange, but it is essentially adjusting package names.\n\nNote that nova did this some time back in change\nIc1fd7c87ceda05eeb96735da2a415ef37060bb1a.\n\nChange-Id: Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25fc8fc3f30b483d4f8a63118c0e532e5fb8b2d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25fc8fc3f30b483d4f8a63118c0e532e5fb8b2d6"}]},"branch":"refs/heads/master"},"5819dd3cfa1c221c80d36d167e56494e0c2a46b6":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2023-02-22 10:29:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/846173/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/846173/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/846173/6"}}},"commit":{"parents":[{"commit":"c70926738284bbf015d3d0390f0db012babea410","subject":"db: Move helpers to cinder.db.sqlalchemy.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c70926738284bbf015d3d0390f0db012babea410"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-15 11:42:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-22 10:29:09.000000000","tz":0},"subject":"db: Use cinder.db.api directly","message":"db: Use cinder.db.api directly\n\ncinder/db/__init__.py was importing * from cinder.db.api. This meant that\nany time any code anywhere within the cinder.db package was imported\nthen cinder.db.api was too, leading to a cascade of imports that may\nnot have been desired. Also, in general, code in __init__.py is a pain.\n\nTherefore, this change adjusts code that so that either:\n\n* cinder.db.api is used directly, or\n* cinder.db.api is imported as \u0027db\u0027\n\nIn either case, the functionality remains the same.\n\nThe primary goal of this change was to make it possible to import the model\nfiles without having to import the DB API. This looks like a very large\nchange, but it is essentially adjusting package names.\n\nNote that nova did this some time back in change\nIc1fd7c87ceda05eeb96735da2a415ef37060bb1a.\n\nChange-Id: Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5819dd3cfa1c221c80d36d167e56494e0c2a46b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5819dd3cfa1c221c80d36d167e56494e0c2a46b6"}]},"branch":"refs/heads/master"},"85c6436e6884a4a1bdab5e72513b75bf6e6f910b":{"kind":"REWORK","_number":7,"created":"2023-02-22 18:52:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/846173/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/846173/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/846173/7"}}},"commit":{"parents":[{"commit":"c70926738284bbf015d3d0390f0db012babea410","subject":"db: Move helpers to cinder.db.sqlalchemy.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c70926738284bbf015d3d0390f0db012babea410"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-15 11:42:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-22 18:52:46.000000000","tz":0},"subject":"db: Use cinder.db.api directly","message":"db: Use cinder.db.api directly\n\ncinder/db/__init__.py was importing * from cinder.db.api. This meant that\nany time any code anywhere within the cinder.db package was imported\nthen cinder.db.api was too, leading to a cascade of imports that may\nnot have been desired. Also, in general, code in __init__.py is a pain.\n\nTherefore, this change adjusts code that so that either:\n\n* cinder.db.api is used directly, or\n* cinder.db.api is imported as \u0027db\u0027\n\nIn either case, the functionality remains the same.\n\nThe primary goal of this change was to make it possible to import the model\nfiles without having to import the DB API. This looks like a very large\nchange, but it is essentially adjusting package names.\n\nNote that nova did this some time back in change\nIc1fd7c87ceda05eeb96735da2a415ef37060bb1a.\n\nChange-Id: Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85c6436e6884a4a1bdab5e72513b75bf6e6f910b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85c6436e6884a4a1bdab5e72513b75bf6e6f910b"}]},"branch":"refs/heads/master"},"56baf866b8445b9625f26214f68023d63266daf4":{"kind":"REWORK","_number":8,"created":"2023-03-30 16:15:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/846173/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/846173/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/846173/8"}}},"commit":{"parents":[{"commit":"5ffd1e8e36f7644dc2a997d0b8be7f9ab38c6ee9","subject":"db: Move helpers to cinder.db.sqlalchemy.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ffd1e8e36f7644dc2a997d0b8be7f9ab38c6ee9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-15 11:42:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-03-30 15:56:59.000000000","tz":60},"subject":"db: Use cinder.db.api directly","message":"db: Use cinder.db.api directly\n\ncinder/db/__init__.py was importing * from cinder.db.api. This meant that\nany time any code anywhere within the cinder.db package was imported\nthen cinder.db.api was too, leading to a cascade of imports that may\nnot have been desired. Also, in general, code in __init__.py is a pain.\n\nTherefore, this change adjusts code that so that either:\n\n* cinder.db.api is used directly, or\n* cinder.db.api is imported as \u0027db\u0027\n\nIn either case, the functionality remains the same.\n\nThe primary goal of this change was to make it possible to import the model\nfiles without having to import the DB API. This looks like a very large\nchange, but it is essentially adjusting package names.\n\nNote that nova did this some time back in change\nIc1fd7c87ceda05eeb96735da2a415ef37060bb1a.\n\nChange-Id: Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56baf866b8445b9625f26214f68023d63266daf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56baf866b8445b9625f26214f68023d63266daf4"}]},"branch":"refs/heads/master"},"83781b9149efdf69acbe70aa78baf254bdd4a1ed":{"kind":"REWORK","_number":9,"created":"2024-11-22 06:09:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/846173/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/846173/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/846173/9"}}},"commit":{"parents":[{"commit":"201946e8321abac4f1730b479c7d6f24e1c2d268","subject":"db: Move helpers to cinder.db.sqlalchemy.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/201946e8321abac4f1730b479c7d6f24e1c2d268"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-15 11:42:17.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 06:08:54.000000000","tz":540},"subject":"db: Use cinder.db.api directly","message":"db: Use cinder.db.api directly\n\ncinder/db/__init__.py was importing * from cinder.db.api. This meant that\nany time any code anywhere within the cinder.db package was imported\nthen cinder.db.api was too, leading to a cascade of imports that may\nnot have been desired. Also, in general, code in __init__.py is a pain.\n\nTherefore, this change adjusts code that so that either:\n\n* cinder.db.api is used directly, or\n* cinder.db.api is imported as \u0027db\u0027\n\nIn either case, the functionality remains the same.\n\nThe primary goal of this change was to make it possible to import the model\nfiles without having to import the DB API. This looks like a very large\nchange, but it is essentially adjusting package names.\n\nNote that nova did this some time back in change\nIc1fd7c87ceda05eeb96735da2a415ef37060bb1a.\n\nChange-Id: Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/83781b9149efdf69acbe70aa78baf254bdd4a1ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/83781b9149efdf69acbe70aa78baf254bdd4a1ed"}]},"branch":"refs/heads/master"},"dd283babd1b826da22fb4c70bc352f18c9e91f8c":{"kind":"REWORK","_number":10,"created":"2024-11-22 06:17:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/846173/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/846173/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/846173/10"}}},"commit":{"parents":[{"commit":"201946e8321abac4f1730b479c7d6f24e1c2d268","subject":"db: Move helpers to cinder.db.sqlalchemy.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/201946e8321abac4f1730b479c7d6f24e1c2d268"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-15 11:42:17.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 06:16:12.000000000","tz":540},"subject":"db: Use cinder.db.api directly","message":"db: Use cinder.db.api directly\n\ncinder/db/__init__.py was importing * from cinder.db.api. This meant that\nany time any code anywhere within the cinder.db package was imported\nthen cinder.db.api was too, leading to a cascade of imports that may\nnot have been desired. Also, in general, code in __init__.py is a pain.\n\nTherefore, this change adjusts code that so that either:\n\n* cinder.db.api is used directly, or\n* cinder.db.api is imported as \u0027db\u0027\n\nIn either case, the functionality remains the same.\n\nThe primary goal of this change was to make it possible to import the model\nfiles without having to import the DB API. This looks like a very large\nchange, but it is essentially adjusting package names.\n\nNote that nova did this some time back in change\nIc1fd7c87ceda05eeb96735da2a415ef37060bb1a.\n\nChange-Id: Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd283babd1b826da22fb4c70bc352f18c9e91f8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd283babd1b826da22fb4c70bc352f18c9e91f8c"}]},"branch":"refs/heads/master"},"5a43390b8e4f345855d6e936323cf5a09b80abbb":{"kind":"REWORK","_number":11,"created":"2025-10-24 13:05:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/846173/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/846173/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/846173/11"}}},"commit":{"parents":[{"commit":"ed0cf44ec791b1e04c248176d28bbaf7041609fd","subject":"db: Move helpers to cinder.db.sqlalchemy.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ed0cf44ec791b1e04c248176d28bbaf7041609fd"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-15 11:42:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 13:04:44.000000000","tz":60},"subject":"db: Use cinder.db.api directly","message":"db: Use cinder.db.api directly\n\ncinder/db/__init__.py was importing * from cinder.db.api. This meant that\nany time any code anywhere within the cinder.db package was imported\nthen cinder.db.api was too, leading to a cascade of imports that may\nnot have been desired. Also, in general, code in __init__.py is a pain.\n\nTherefore, this change adjusts code that so that either:\n\n* cinder.db.api is used directly, or\n* cinder.db.api is imported as \u0027db\u0027\n\nIn either case, the functionality remains the same.\n\nThe primary goal of this change was to make it possible to import the model\nfiles without having to import the DB API. This looks like a very large\nchange, but it is essentially adjusting package names.\n\nNote that nova did this some time back in change\nIc1fd7c87ceda05eeb96735da2a415ef37060bb1a.\n\nChange-Id: Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5a43390b8e4f345855d6e936323cf5a09b80abbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5a43390b8e4f345855d6e936323cf5a09b80abbb"}]},"branch":"refs/heads/master"},"fc41eae3c123e6bf6e521cd6534935dad2ba618e":{"kind":"REWORK","_number":12,"created":"2026-05-12 13:42:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/846173/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/846173/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/846173/12"}}},"commit":{"parents":[{"commit":"f61226f770b92a01868b79788818c40ff41d8859","subject":"db: Move helpers to cinder.db.sqlalchemy.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f61226f770b92a01868b79788818c40ff41d8859"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-15 11:42:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-12 12:40:26.000000000","tz":60},"subject":"db: Use cinder.db.api directly","message":"db: Use cinder.db.api directly\n\n`cinder/db/__init__.py` was importing `*` from `cinder.db.api`. This\nmeant that any time any code anywhere within the `cinder.db` package was\nimported then `cinder.db.api` was too, leading to a cascade of imports\nthat may not have been desired. Also, in general, code in `__init__.py`\nis a pain.\n\nTherefore, this change adjusts code that so that either:\n\n* `cinder.db.api` is used directly, or\n* `cinder.db.api` is imported as `db`\n\nIn either case, the functionality remains the same.\n\nThe primary goal of this change is to make it possible to import the\nmodel files without having to import the DB API. This looks like a very\nlarge change, but it is essentially adjusting package names.\n\nNote that nova did this some time back in change\nIc1fd7c87ceda05eeb96735da2a415ef37060bb1a.\n\nChange-Id: Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fc41eae3c123e6bf6e521cd6534935dad2ba618e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fc41eae3c123e6bf6e521cd6534935dad2ba618e"}]},"branch":"refs/heads/master"},"43564068e28905683750522d9fed4f14bf5fa580":{"kind":"REWORK","_number":13,"created":"2026-05-12 15:42:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/846173/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/73/846173/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/73/846173/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/73/846173/13"}}},"commit":{"parents":[{"commit":"cf43363c37e70faef2cd8f2aaf5c9213f3df5b49","subject":"db: Move helpers to cinder.db.sqlalchemy.api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf43363c37e70faef2cd8f2aaf5c9213f3df5b49"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-15 11:42:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-12 15:12:56.000000000","tz":60},"subject":"db: Use cinder.db.api directly","message":"db: Use cinder.db.api directly\n\n`cinder/db/__init__.py` was importing `*` from `cinder.db.api`. This\nmeant that any time any code anywhere within the `cinder.db` package was\nimported then `cinder.db.api` was too, leading to a cascade of imports\nthat may not have been desired. Also, in general, code in `__init__.py`\nis a pain.\n\nTherefore, this change adjusts code that so that either:\n\n* `cinder.db.api` is used directly, or\n* `cinder.db.api` is imported as `db`\n\nIn either case, the functionality remains the same.\n\nThe primary goal of this change is to make it possible to import the\nmodel files without having to import the DB API. This looks like a very\nlarge change, but it is essentially adjusting package names.\n\nNote that nova did this some time back in change\nIc1fd7c87ceda05eeb96735da2a415ef37060bb1a.\n\nChange-Id: Ie73c74c0eeeda185a8d648444a88f7ddcbdf83b4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43564068e28905683750522d9fed4f14bf5fa580"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43564068e28905683750522d9fed4f14bf5fa580"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
