)]}'
{"id":"openstack%2Fcinder~846230","triplet_id":"openstack%2Fcinder~master~I21a3df40759ec57e2fe27fe31816ee14457847c8","project":"openstack/cinder","branch":"master","topic":"datacore-ci-test","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-06-17 19:26:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I21a3df40759ec57e2fe27fe31816ee14457847c8","subject":"DNM: CI test probe","status":"ABANDONED","created":"2022-06-16 22:05:06.000000000","updated":"2022-06-20 05:26:33.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c2abde3905be85ba252207392b0dbf6bef51ed69","_number":846230,"virtual_id_number":846230,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-17 00:16:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34791,"name":"Arun KV","email":"arun.kv@datacore.com","username":"dcarun"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-16 22:28:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-16 22:57:21.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-06-16 23:13:09.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-06-17 00:16:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-17 01:16:48.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-17 01:33:44.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-17 02:26:16.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-06-17 03:55:01.000000000","updated_by":{"_account_id":34791,"name":"Arun KV","email":"arun.kv@datacore.com","username":"dcarun"},"reviewer":{"_account_id":34791,"name":"Arun KV","email":"arun.kv@datacore.com","username":"dcarun"},"state":"CC"},{"updated":"2022-06-17 03:58:32.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-17 04:40:03.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-17 08:23:55.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-06-17 08:53:42.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-17 16:24:24.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-06-20 05:26:33.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"3fe04f3939d3418b96d263af68b4536ad8b2f903","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-06-16 22:05:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eab074c7ba8939d93c365e29ba379f8abfa65fc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 22:28:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/68e4356c439b4ed7ad43f47bbf873dc9 : FAILURE in 21m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92a068db184b4ffca0295773c2631468 : FAILURE in 15m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d264ffe31c774a898dbb2a51f7d2e0ed76a42a8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-16 22:57:21.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1655417115186 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"7d786690e0cfbfd1c9101cc3a718a81705bd470b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-16 23:13:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/846230/1/iscsi/ : SUCCESS in 1h 07m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/846230/1/smb/ : SUCCESS in 1h 05m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9db37f71d66c6c5002c35478dfe231544e5053c8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-06-16 23:30:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b683be151187d755536050f3b01206eba03f4fe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 00:16:04.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8bdffab7ed748bd9b8353069b1b5b2c : FAILURE in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6f79257e8f74a59a16de5dda830552a : FAILURE in 11m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccfc4a729e64498cb600731c2605e78d : FAILURE in 12m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5ca3a6d992c4c81ad3deb6521f0b4d0 : FAILURE in 9m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/030ea04bc27a4e77a1e7b6fbbe5bffde : SUCCESS in 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/91ef53fe28fc41fda172eb9ccf3c9093 : SUCCESS in 55m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/72b8a14254cf4e198ba94faec350ccb7 : SUCCESS in 1h 10m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44ab084ecd404e68b70d61c3892517c7 : SUCCESS in 45m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/869490e8bcc04c7f92fca2c48a383a65 : FAILURE in 12m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/01b1395b7e37465da5f0ccd32adb25ec : SUCCESS in 4m 54s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/975e31213c8745e68247a3ec3b27eea8 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e217515c742b49cf9113da8aaf080f75 : SUCCESS in 6m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b6f8de170a9f4749ad76a3eb15311a87 : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2b2b629596d849c0a1b6292207d52e6b : SUCCESS in 46m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/16821e22d66444f8882cf45ce6f4af6c : FAILURE in 6m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dce9e83bb27e4e3a806c7d0cda24daa2 : SUCCESS in 1h 09m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/42a8c05cb2cc465fbeadf49761e43738 : FAILURE in 1h 16m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3990a8d1a53148408c0d4e540b9ca507 : SUCCESS in 1h 13m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/be25939e91344605a899c20d77cbabd7 : SUCCESS in 1h 07m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2cfa032b89a34b3cb2b60165c3f7d01f : SUCCESS in 1h 15m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3e08cd6d8cba44918fd5d20f6551f014 : SUCCESS in 43m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/06230141f37a42f8ab4f8dd5ffeb8340 : FAILURE in 59m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/02de1c69646843b28adf02d1a3b47592 : SUCCESS in 2h 00m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9a74ca4085242329244f854b3d058c9 : SUCCESS in 49m 03s","accounts_in_message":[],"_revision_number":1},{"id":"38ce6f744b2d81eb1c2ac864fa772f5afc0a69d6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-17 01:16:48.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/44ee6ad6f437440aa5db5a54078d19c2 : POST_FAILURE in 1h 41m 03s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/e2907cbd931d4684abcc9cdd98c8c42a : TIMED_OUT in 3h 05m 22s","accounts_in_message":[],"_revision_number":1},{"id":"256ce1480731f298b0e8dc636e131c3dd5978153","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-17 01:33:44.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-30-846230-1/ns5_nfs : SUCCESS in 3h 28m 8s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-30-846230-1/ns5_iscsi : SUCCESS in 3h 25m 28s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"b0b94201b69f219482843f36489d475cbf2491df","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-17 02:26:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k-focal https://hpe-storage.github.io/alletra-6k-ci-logs/logs/30/846230/1/check/iscsi-driver-alletra-6k-focal/5354f4a/5354f4a.html : FAILURE in 51m 34s","accounts_in_message":[],"_revision_number":1},{"id":"20fd7f4f8e073d3e3851c41b21c1f9a9b13d6e3c","author":{"_account_id":34791,"name":"Arun KV","email":"arun.kv@datacore.com","username":"dcarun"},"date":"2022-06-17 03:55:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f18ef0ea92baeb8e38e62e5659d68cabbd005ae5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-17 03:55:28.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1655438110880 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"8cf504643f77a944941e0e4a37ff948c427a2979","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-06-17 03:58:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/846230 : FAILURE in 1h 27m 16s","accounts_in_message":[],"_revision_number":1},{"id":"7525614f3de3bee55354ceda874c3bf4e536e91f","author":{"_account_id":34791,"name":"Arun KV","email":"arun.kv@datacore.com","username":"dcarun"},"date":"2022-06-17 04:01:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d378406bc40b6fcc002159487c30ffb196c62803","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-17 04:02:12.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1655438515887 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ed07e2f4b41326a1ee182ca5fb41918e2c2dfb4c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-17 04:40:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3c58102079274591aa11a7191c031468 : SUCCESS in 38m 01s","accounts_in_message":[],"_revision_number":1},{"id":"de0eed31414b8371f923d2fa3187afa9e895c95f","author":{"_account_id":34791,"name":"Arun KV","email":"arun.kv@datacore.com","username":"dcarun"},"date":"2022-06-17 05:37:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd86856a27dea1d239c58153e7dad434ffa28fbe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-17 05:42:39.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1655444251076 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"c64177693cf9cf6d46199dd67b22709c66ffee80","author":{"_account_id":34791,"name":"Arun KV","email":"arun.kv@datacore.com","username":"dcarun"},"date":"2022-06-17 06:32:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca60aca6375eedd3853ac81d00bf11e05056c643","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-17 06:37:44.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1655447561190 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ddfd13ee0fb28b20dd1b2b2d4254bff55eabbf88","author":{"_account_id":34791,"name":"Arun KV","email":"arun.kv@datacore.com","username":"dcarun"},"date":"2022-06-17 06:42:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74325b4305e61e963f72ee652341c3cef637fbdf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-17 06:48:13.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1655448181221 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"ec074c4495487bd396b7f1810751a63e593ce08f","author":{"_account_id":34791,"name":"Arun KV","email":"arun.kv@datacore.com","username":"dcarun"},"date":"2022-06-17 08:19:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ede8ffd2c2988521daa2648076100e525e2b8c31","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-17 08:23:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/30/846230/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f23442c/ : TIMED_OUT in 3h 02m 54s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/30/846230/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/1a4133f/ : FAILURE in 1h 07m 49s","accounts_in_message":[],"_revision_number":1},{"id":"42d82765537739fe812a058f415aacb043329016","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-17 08:24:44.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1655454006386 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"cfe8e004e6c788858564e8f644ef0f4fc3a935fd","author":{"_account_id":34791,"name":"Arun KV","email":"arun.kv@datacore.com","username":"dcarun"},"date":"2022-06-17 08:27:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36e37660238b2cdff53b439b5de085599c1387c3","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-06-17 08:53:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/30/846230/1/ : RETRY_LIMIT in 8s","accounts_in_message":[],"_revision_number":1},{"id":"f0c232950c45ff805d470da7362efe7fe87e9527","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-17 08:54:00.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1655454441436 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"a95cd4d5a375ccf2c827e592d91944de317b269d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-17 11:52:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/30/846230/1/check/iscsi-driver-nimble/06ff229/06ff229.html : FAILURE in 47m 33s","accounts_in_message":[],"_revision_number":1},{"id":"80adaab04a3a22f341388ae4a6b4211a52a66636","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-17 16:24:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/30/846230/1/check/dsvm-tempest-cisco-zonemanager-job/1e69e93/ : SUCCESS in 1h 09m 01s","accounts_in_message":[],"_revision_number":1},{"id":"422ccc49a31b23f0cc8648d258309b0f5872a0af","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-06-17 19:26:56.000000000","message":"Abandoned\n\nNo longer needed.","accounts_in_message":[],"_revision_number":1},{"id":"c2abde3905be85ba252207392b0dbf6bef51ed69","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-20 05:26:33.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae10524593b823a6a9b5fa16d895301d2e81d5e0","revisions":{"ae10524593b823a6a9b5fa16d895301d2e81d5e0":{"kind":"REWORK","_number":1,"created":"2022-06-16 22:05:06.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/30/846230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/30/846230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/846230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/846230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/30/846230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/30/846230/1"}}},"commit":{"parents":[{"commit":"0f066efb691db5e95d11372328944d0311a16acb","subject":"Reintroduce DataCore driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0f066efb691db5e95d11372328944d0311a16acb"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-06-16 22:04:35.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-06-16 22:04:35.000000000","tz":-240},"subject":"DNM: CI test probe","message":"DNM: CI test probe\n\nChange-Id: I21a3df40759ec57e2fe27fe31816ee14457847c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae10524593b823a6a9b5fa16d895301d2e81d5e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae10524593b823a6a9b5fa16d895301d2e81d5e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
