)]}'
{"id":"openstack%2Fcinder~847131","triplet_id":"openstack%2Fcinder~master~Idbac4dae00f9fa03fb14547a204666721bf67d96","project":"openstack/cinder","branch":"master","topic":"bp/clone-across-pools","attention_set":{"12988":{"account":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"last_update":"2024-07-31 18:48:20.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2024-07-25 02:53:01.000000000","reason":"Change was submitted"},"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-07-25 02:53:01.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-07-17 13:32:20.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2024-07-25 02:53:01.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-07-25 02:53:01.000000000","reason":"Change was submitted"},"33612":{"account":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"last_update":"2024-07-24 06:31:11.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_29122\u003e by clicking the attention icon","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"7198":{"account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"last_update":"2024-07-24 14:15:21.000000000","reason":"\u003cGERRIT_ACCOUNT_7198\u003e replied on the change","reason_account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-07-25 02:53:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idbac4dae00f9fa03fb14547a204666721bf67d96","subject":"Add the clone_across_pools driver capability","status":"MERGED","created":"2022-06-22 07:58:43.000000000","updated":"2024-11-01 04:00:08.000000000","submitted":"2024-07-25 02:53:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":38,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"847131-bp/clone-across-pools","meta_rev_id":"2249c72d3e6acafe7bb383cafa35b85588ba9031","_number":847131,"virtual_id_number":847131,"owner":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-25 02:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":2,"date":"2024-07-25 01:09:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-24 14:15:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2024-07-25 01:09:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37093,"name":"ceastor_cinder_ci","display_name":"ceastor_cinder_ci","email":"openstack_ci@163.com","username":"ceastor_cinder_ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-22 08:09:29.000000000","updated_by":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-06-22 08:18:36.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-06-22 08:31:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-22 10:23:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-22 10:23:12.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-06-22 10:28:40.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-06-22 11:07:22.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-06-22 12:03:22.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-06-22 12:25:47.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-06-22 12:32:47.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-22 12:42:10.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-22 13:34:26.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-22 14:32:01.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-22 14:35:13.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-06-22 15:33:32.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-06-22 15:54:48.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-06-22 17:24:58.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-06-22 20:10:02.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-06-22 21:30:38.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-06-22 21:56:27.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-06-23 04:57:20.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-06-23 08:02:38.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-23 17:09:14.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-15 18:07:04.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-07-19 14:11:35.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-01 14:48:51.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-01 16:19:53.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-01 20:05:57.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-08-02 12:04:09.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-01-17 14:22:30.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-24 10:48:19.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-01-24 19:05:05.000000000","updated_by":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"reviewer":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"state":"CC"},{"updated":"2023-02-08 13:54:12.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-13 10:25:45.000000000","updated_by":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"reviewer":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"state":"CC"},{"updated":"2023-02-14 03:08:48.000000000","updated_by":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"reviewer":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"state":"REMOVED"},{"updated":"2023-05-08 22:18:54.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-07-21 13:23:35.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2023-08-01 21:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-07-08 16:59:15.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"CC"},{"updated":"2024-07-08 16:59:15.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-07-08 16:59:15.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-07-08 17:22:04.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-07-09 07:59:44.000000000","updated_by":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"reviewer":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"state":"CC"},{"updated":"2024-07-09 12:35:10.000000000","updated_by":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"reviewer":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"state":"CC"},{"updated":"2024-07-18 05:30:38.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-07-18 05:30:42.000000000","updated_by":{"_account_id":37093,"name":"ceastor_cinder_ci","display_name":"ceastor_cinder_ci","email":"openstack_ci@163.com","username":"ceastor_cinder_ci"},"reviewer":{"_account_id":37093,"name":"ceastor_cinder_ci","display_name":"ceastor_cinder_ci","email":"openstack_ci@163.com","username":"ceastor_cinder_ci"},"state":"CC"},{"updated":"2024-07-18 16:27:01.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2024-07-24 06:29:46.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"}],"messages":[{"id":"ab3912d2f4e98fc170042443de3e2190f5d03303","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2022-06-22 07:58:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29ed25a97becd4b862cfcf704c309a9233b2bdaa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-22 08:18:36.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1655884733140 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"2fa7dbb3f8bff41ee16ebcf3d4419cea7754c05d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 08:31:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0189534f4d614e11929800ccbd23b38a : FAILURE in 23m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a4043469195473cb2f82ff11815e873 : SUCCESS in 23m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d5f25e00ebaa3356549bd9d0fd671891068bd9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 10:23:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7d8f200f5984a1c9ba8fb9224cff721 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb390797840349db9fec088dedeed71a : SUCCESS in 13m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44d23518b858412b9b7ad256c65ef02a : SUCCESS in 12m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69c214e69cbb4671a4f867744cee63dd : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7a31457acf94a199448bd2869386cac : SUCCESS in 11m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/1210761182424261b5ee5229d9f7a7f1 : SUCCESS in 1h 06m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b56044711b9f48b89c23d75566ff656f : SUCCESS in 1h 20m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29628ce833a74050af37ece477976c94 : SUCCESS in 1h 00m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/810963a1c2934c23a23e028265b6c7e5 : SUCCESS in 15m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b8a5563453384b269f539158d1ddebbb : SUCCESS in 5m 33s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6ec0e02bdf5c4b4e98e2a22ff2915dd2 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a13c186cd09245c182286f100b22fc5b : SUCCESS in 6m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0e5ea5063ff94f528e7f8c70ea5d6811 : SUCCESS in 7m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ac5783d91d3f4770bae838a55a38b5d7 : SUCCESS in 46m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8851cbcb6a0b4abb911d9fca47de77e4 : FAILURE in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e469cc17547d41d5a4afa5f782f8b4ef : SUCCESS in 1h 16m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/747e330979bb4c16a82268afeed55d8f : FAILURE in 1h 48m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cb10676b4f324b11a45a011aacc717ee : SUCCESS in 1h 43m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8bc423fd4e2f4cfd931e2ca246b7aeb1 : SUCCESS in 1h 06m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/22167174a1c14c8883e07cfa2a875b03 : SUCCESS in 1h 37m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b1d542a417aa483bb29e91b437757f34 : SUCCESS in 54m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5c67b132c2204fc8b0d51010a252b093 : FAILURE in 1h 15m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/49317a0165d347f3bb96ce29f014875b : SUCCESS in 2h 18m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5a3a98b529f4fe1af2223e94fee0dfc : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a0b7fbb382d193201a4fe2908e3c27fee1fbb46d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-06-22 10:23:12.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"3a382c44c04168a1c8d5f44c2a37671dd0f67bf5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-22 10:28:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/847131/1/iscsi/ : SUCCESS in 1h 03m 37s\n- cinder-smb http://cloudbase-ci.com/cinder/847131/1/smb/ : SUCCESS in 58m 02s","accounts_in_message":[],"_revision_number":1},{"id":"fd423969ae01e89ed9cdf3b8d556042c9aecf62c","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-06-22 11:07:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-31-847131-1/1655893385 : SUCCESS in 16m 43s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":1},{"id":"37492711c3d56060e5675ca477e3b192f9f03dd1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-22 12:03:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k-focal https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/847131/1/check/iscsi-driver-alletra-6k-focal/fefe1ed/fefe1ed.html : FAILURE in 1h 30m 09s","accounts_in_message":[],"_revision_number":1},{"id":"40846a55a47b640b9d5a93f7f4297a5969a4e712","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-22 12:25:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/1/upstream-check/cinder-tempest-plugin-ontap-fcp/850e08a/ : SUCCESS in 2h 00m 54s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/dc7601c/ : SUCCESS in 43m 14s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/1/upstream-check/cinder-tempest-plugin-ontap-nfs/a677314/ : SUCCESS in 47m 24s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"6d17b4bf221cfedbffee3493efda5857d1ee154d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-22 12:32:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7ed8957bdd6749bc93157d479f15394f : SUCCESS in 40m 45s","accounts_in_message":[],"_revision_number":1},{"id":"beb85010f9ab99bec0dcd91012449dcc0286823e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-22 12:42:10.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/c806b75daede4d8eb99e00cc1378a0a6 : POST_FAILURE in 3h 03m 14s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/7bd0917b3a6e465f8efe824c0f2beb61 : POST_FAILURE in 2h 39m 15s","accounts_in_message":[],"_revision_number":1},{"id":"ce30598df8a5332b866eb69a93ca897535cf93d9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-22 13:34:26.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-847131-1/ns5_nfs : SUCCESS in 3h 36m 6s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-847131-1/ns5_iscsi : SUCCESS in 3h 42m 19s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"6c4a6d6c08cfce7d28e3f4772bc686d39a35aa49","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-06-22 14:32:01.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/31/847131/1/check/DellEMC_PowerFlex/4b81a6a/DellEMC_PowerFlex/884 : SUCCESS in 1h 24m 24s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"61d2ad04768a3ced60a9a3b65a0453863d190666","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-22 14:35:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/847131/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/9bd51c4/ : TIMED_OUT in 3h 17m 30s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/847131/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ce42ea3/ : FAILURE in 1h 20m 13s","accounts_in_message":[],"_revision_number":1},{"id":"84c39f342dfb80096c7423068ba7245600398be1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-22 15:24:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/31/847131/1/check/iscsi-driver-nimble/6f75e4c/6f75e4c.html : FAILURE in 1h 32m 05s","accounts_in_message":[],"_revision_number":1},{"id":"121b88a412d5e697eaf25d4c6781b44816c5885c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-22 15:33:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/31/847131/1/check/dsvm-tempest-cisco-zonemanager-job/3dab457/ : SUCCESS in 1h 08m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c717018ff0a30c481de9120b182933c7dd287900","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-06-22 15:54:48.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/31/847131/1/check/DellEMC_SC_FC/0640121/DellEMC_SC_FC/730 : SUCCESS in 51m 08s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/1/check/DellEMC_SC_iSCSI/9048bf9/DellEMC_SC_iSCSI/752 : SUCCESS in 1h 45m 18s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"098a5394e2ace2a6476a29cc3743cd533dfb93f3","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-06-22 17:24:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/847131-1/HPMSA-iSCSI/ : FAILURE in 53:52.81\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/847131-1/HPMSA-FC/ : FAILURE in 48:56.88","accounts_in_message":[],"_revision_number":1},{"id":"ad10485a915a7aa3a2f4bfbf3afdcd12b9942afa","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-06-22 20:10:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/847131-1/PVME-iSCSI/ : FAILURE in 1:05:50\n- PVME-FC https://os-logs.seagate.com/dell-ci/847131-1/PVME-FC/ : FAILURE in 1:02:32","accounts_in_message":[],"_revision_number":1},{"id":"236bcc23b620903b69b4305994d612f8eb362ea7","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-06-22 21:30:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/847131-1/Lenovo-iSCSI/ : FAILURE in 1:06:04\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/847131-1/Lenovo-FC/ : FAILURE in 1:01:16","accounts_in_message":[],"_revision_number":1},{"id":"1fa17ac66378c1bec51704296b2935a82d988117","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-06-22 21:56:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/847131-1/Seagate-iSCSI/ : FAILURE in 1:07:31\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/847131-1/Seagate-FC/ : FAILURE in 1:05:32","accounts_in_message":[],"_revision_number":1},{"id":"2aeedbb97c33524d913154bb048a3d46985ae38b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-23 04:57:20.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/1/check/DellEMC_Unity_FC/00c1c41/DellEMC_Unity_FC/1000 : SUCCESS in 1h 12m 06s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/1/check/DellEMC_Unity_iSCSI/f3f7a19/DellEMC_Unity_iSCSI/940 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"db0708f8ed39f405fa385f3c711e8fd18970e410","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-23 08:02:38.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"c68bae7dca1f4050dcb7be04d729de879220dfa3","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-06-23 17:09:14.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/31/847131/1/check/DellEMC_PowerMAX_FC/de15c8e/DellEMC_PowerMAX_FC/605 : SUCCESS in 3h 26m 48s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/1/check/DellEMC_PowerMAX_iSCSI/88ca86e/DellEMC_PowerMAX_iSCSI/592 : SUCCESS in 5h 05m 12s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"dcfba7fe6c5460ab3ba7488415d67686eb6551d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2022-06-24 07:50:17.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"cbc0386248d7de3d9b8d666c7a58fa802df1052a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 08:32:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8333bca5cabc4bc8aa4336fcefaaa576 : SUCCESS in 16m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/beab62c297c64dd7969e60140df9014d : FAILURE in 30m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9a3ec8bfe14ad847532c9fabc42f2b254f0e5d4b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-24 09:12:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/847131/2/iscsi/ : SUCCESS in 1h 05m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/847131/2/smb/ : SUCCESS in 1h 21m 17s","accounts_in_message":[],"_revision_number":2},{"id":"f10af3cd9495675146d45f794e1ac201bc2e8a63","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-24 10:07:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b262f94b126c46e7aa7fa14119137513 : FAILURE in 1h 23m 22s","accounts_in_message":[],"_revision_number":2},{"id":"a3517f53be4b026ae41589539ef14396faf3f92f","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2022-06-24 10:09:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"034f1932ac2c6ad1d6802a931fd56d06070a0eec","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-24 10:10:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/847131/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/454f4a4/ : SUCCESS in 1h 24m 54s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/847131/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/56b85fe/ : SUCCESS in 1h 22m 14s","accounts_in_message":[],"_revision_number":2},{"id":"e5d5ec3b48b90753cafefbfc9c474bb331abcd88","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2022-06-24 10:12:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"37df997ad00276a970ef188c3724fd9538a97588","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 10:27:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/220f7aff27af4f26829293575258a602 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ed0864ef6f947dfb7eab9050d6e3fd4 : SUCCESS in 12m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a5912b3d55f44f982895d26ea397410 : FAILURE in 12m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94705d0a65354edf85ecc8e367ad1f28 : SUCCESS in 11m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/192fd07d145240feb38e84f1b887cbd7 : SUCCESS in 12m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/9faf8d66db9740a6b2bf70bc9909ce2a : SUCCESS in 1h 05m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b1b08679a06e412ca02e8d962e50f126 : SUCCESS in 1h 17m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e671b217b08e47e8a84e819f77c9a416 : SUCCESS in 50m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fdd47f1d25264a87b9c96710bcdb872a : SUCCESS in 14m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f5f4e68c16c24dfb8960f91f7443f2b9 : SUCCESS in 5m 30s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/571d0678f4434ce8a642dd0521f744fb : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/86b6686c2ff74abdae8062f4fb6590b1 : SUCCESS in 6m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fe42ded274d042749d24ff2e77e9f601 : SUCCESS in 6m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5471926de6154de68341f860cf2e6c24 : SUCCESS in 44m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2a848f96eca54bde929e34b9e170b753 : FAILURE in 6m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/af641983ef48472296a5207b66420093 : SUCCESS in 1h 15m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/78af8d46e1944361b0957d8d5c215580 : FAILURE in 1h 23m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/db613c5db88c43dd93783a26b5c138a2 : SUCCESS in 1h 54m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cb4714d54dc94f5b9d962d1c8efe7661 : SUCCESS in 1h 03m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9a062b7b6efb44bd8776e550ea2e88fb : SUCCESS in 1h 34m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/716d38e25e884e539eb2011082fd8f69 : SUCCESS in 44m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b5541873acc94e658fc8562f43a5ac10 : FAILURE in 1h 02m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d399b8c12c754fb5b6f8ce58a451de6d : SUCCESS in 2h 29m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6058c766182498e992c64843fb9c64e : SUCCESS in 51m 33s","accounts_in_message":[],"_revision_number":2},{"id":"bfbe5f67c575d33379fec41f9a428360dd193ec4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-24 10:55:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/847131/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c2c0965/ : TIMED_OUT in 3h 02m 55s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/847131/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/5f4623b/ : FAILURE in 1h 38m 52s","accounts_in_message":[],"_revision_number":2},{"id":"379a26e6e9b0e4e8bd5789e8af2a5d37d1ee72fc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-24 11:03:39.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/31/847131/2/check/tempest-dsvm-full-focal-py3/5c97ed3/ : TIMED_OUT in 3h 10m 18s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/31/847131/2/check/devstack-plugin-ceph-dsvm-focal-py3/97f73af/ : TIMED_OUT in 3h 08m 25s","accounts_in_message":[],"_revision_number":2},{"id":"93a55afb3d4e04903f7061105d666398e63212cf","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-24 11:37:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k-focal https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/847131/2/check/iscsi-driver-alletra-6k-focal/39ddd78/39ddd78.html : FAILURE in 1h 30m 32s","accounts_in_message":[],"_revision_number":2},{"id":"435efefeeb4ccb543a9875504262f260715148f4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-24 11:56:51.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-847131-2/ns5_nfs : SUCCESS in 3h 37m 35s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-847131-2/ns5_iscsi : SUCCESS in 3h 43m 29s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"abdff6ba9f18d1f1502e52a2b73978f8fab74068","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-24 15:45:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/31/847131/2/check/iscsi-driver-nimble/2606308/2606308.html : FAILURE in 1h 31m 45s","accounts_in_message":[],"_revision_number":2},{"id":"fedc56707cc8370c2ea4f2351386f982fd3515f3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-25 02:45:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/31/847131/2/check/dsvm-tempest-cisco-zonemanager-job/a054b0c/ : SUCCESS in 1h 09m 30s","accounts_in_message":[],"_revision_number":2},{"id":"df6e8c3477076f2e7bf83b3db6a4442e659b7432","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-25 08:04:09.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/2/check/DellEMC_Unity_FC/0bbf7b9/DellEMC_Unity_FC/1027 : SUCCESS in 1h 08m 17s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/2/check/DellEMC_Unity_iSCSI/0fdb181/DellEMC_Unity_iSCSI/963 : SUCCESS in 1h 54m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"42933889476d15b7b93b4eb3e4f8aab0aaaeed00","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-15 18:07:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/2/20220715205452/NEC-V-FC : SUCCESS in 1h 55m 22s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/2/20220715205452/NEC-V-ISCSI : SUCCESS in 6h 12m 09s","accounts_in_message":[],"_revision_number":2},{"id":"ab52dd74766bf634f44026f47b2d33a4ec5b3801","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-07-19 14:11:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/2/20220719193948/NEC-FC : SUCCESS in 3h 28m 22s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/2/20220719193948/NEC-ISCSI : SUCCESS in 3h 31m 43s","accounts_in_message":[],"_revision_number":2},{"id":"77822f03d38e00eb58d23f18f416333be2759bf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2022-08-01 13:13:55.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"2c28803770a7e4614674813ec79c5e31cf95a117","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-01 13:16:46.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"0c49b769a985ddd84f87dd7bb59f10e73b608864","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-01 14:24:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/847131/3/iscsi/ : SUCCESS in 1h 09m 06s\n- cinder-smb http://cloudbase-ci.com/cinder/847131/3/smb/ : SUCCESS in 1h 09m 15s","accounts_in_message":[],"_revision_number":3},{"id":"1f535f8b6b0b20865c51c265fcd45ca615a1d95e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-01 14:48:51.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847131 : SUCCESS in 1h 28m 29s","accounts_in_message":[],"_revision_number":3},{"id":"c8175f617d5e804e8b5a11713bf21c9841b4e306","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-01 14:55:10.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/847131/3/check/iscsi-driver-alletra-6k/d1d23ca/d1d23ca.html : FAILURE in 1h 38m 56s","accounts_in_message":[],"_revision_number":3},{"id":"1188e3262bb80339d0d4da7492ce6b4fface0335","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 15:00:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95dc84ca11a3416dbb3f77f44943bee0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e7522f9242ca4f6195940e5bf48c41e0 : FAILURE in 38m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4686561483c4453490cbbde8173db9a2 : SUCCESS in 23m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d8747b029040063d92274d72b41e80a48ffb81e5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-01 15:08:13.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/a26f20d1e8754095824c8a19bf413a73 : POST_FAILURE in 1h 49m 45s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"2a0ca31891a2a9106d1a35f47712c323d025740b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-01 15:08:21.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/847131/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b993b7d/ : SUCCESS in 1h 45m 25s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/847131/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3594dd6/ : SUCCESS in 1h 36m 54s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/847131/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/b5f4822/ : SUCCESS in 1h 43m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"878e553a4ef5f7980ac71a00e14213d6f8901a59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 15:20:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c3e0e05a5014540b182bbdbd3a7debf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18ce202150f24e9abc7e46f7c81d408c : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/708860ed13984ab3a098f07468caf5ea : SUCCESS in 13m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7882418c8ff644b39c08584b6abdf9d2 : TIMED_OUT in 40m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be627ded61b94874a89061ffd689b4a9 : SUCCESS in 13m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0fc46a547264e99a73587a7f8944a4a : FAILURE in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/6c1d5147afe14791b1fbd54ca0ec1a80 : SUCCESS in 1h 07m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/343a93a0e37142a8ae8dcea9679eb9fc : SUCCESS in 1h 29m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/363c95d7a689479198fdbc11910e4e49 : SUCCESS in 1h 01m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8574a4ce519647ee9f54c80a9828a00f : SUCCESS in 17m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e1c0fa2bb8104b81af7b93846d393822 : SUCCESS in 6m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/041aa93ae3c5498782cceefe7dc1ffe5 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f41db9ce2acc47439212de52327f57c7 : SUCCESS in 7m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/43930c0dba6a4da990864014415ed890 : SUCCESS in 7m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6a5ebf64bc2147d5af8227740909f0ee : SUCCESS in 39m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5b05f0766e9d4d3d8e8ed685afa27c9e : FAILURE in 7m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6a8beebb5bf94d9a93c906fa6a6c12b6 : SUCCESS in 1h 19m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9d434ba2463e48058121394a1bf3881b : FAILURE in 1h 56m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f0355ad1538345a69570e36143d550fa : SUCCESS in 1h 53m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f3225c7420f5455abca826de30ea72f8 : SUCCESS in 1h 19m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3b54edc202e44c198235b779c74d223c : SUCCESS in 1h 23m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2625f0d40ab0416895725680948e7481 : SUCCESS in 39m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/87595fa16b8141bbae60497fec24d59b : SUCCESS in 1h 41m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9a976ae6175644e3bee19b2610a9f63b : SUCCESS in 2h 04m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfb4e444d86f4517a77ba2df4e344554 : SUCCESS in 1h 06m 27s","accounts_in_message":[],"_revision_number":3},{"id":"6c030eff9a3fab3f24fe942acc6bf579c044e31f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-08-01 15:43:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/73277afe6ada4805b349aaf755bb74f3 : SUCCESS in 42m 03s","accounts_in_message":[],"_revision_number":3},{"id":"80fb23477626ac659b418a39bae1a9d5f8e88c1d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-01 16:19:53.000000000","message":"Patch Set 3:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/31/847131/3/check/fujitsu-eternus-dx-iscsi/7d70873 : SUCCESS in 1h 35m 40s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/31/847131/3/check/fujitsu-eternus-dx-fc/e5ab69f : SUCCESS in 1h 50m 29s","accounts_in_message":[],"_revision_number":3},{"id":"047a56505d6c490bdd8d7a255a9a28b42b8f296e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-01 16:57:54.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-847131-3/ns5_nfs : SUCCESS in 3h 25m 57s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-847131-3/ns5_iscsi : SUCCESS in 3h 28m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"dce3e1c1451eaf8cc2624483cb76171271ad4c9f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-01 17:17:41.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/3/check/DellEMC_Unity_FC/4231613/DellEMC_Unity_FC/1319 : FAILURE in 19m 16s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/3/check/DellEMC_Unity_iSCSI/8ef90d7/DellEMC_Unity_iSCSI/1199 : SUCCESS in 2h 05m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"6a43fb29454953abbcaedf6ef066e38ca92456c9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-01 18:01:40.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1659359645862 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"fa795df55fb97bb8c56b2345732a8818437a7ebd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-01 20:05:57.000000000","message":"Patch Set 3:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/344","accounts_in_message":[],"_revision_number":3},{"id":"6f4bbbb5c2d493f0a9240cb8de7e078bdbb0e250","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2022-08-02 08:04:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d6d280d954a456d77f9fec346becdf518ea0e0c0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-02 08:06:30.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"b5a45eddfdc82c8d0fa0a51993fe24ffb15aa186","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 08:28:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28ff8872bb544e95be7f93bb95c6c1bb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a253298d7aaf408495c06b24c1433ea2 : SUCCESS in 21m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12130c9c30ba4789be6aecdac8254d7c : SUCCESS in 21m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b369f882b9dc07be5109a5937e2f8ff0cb883916","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-08-02 08:38:15.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/3/20220802150501/NEC-V-FC : SUCCESS in 1h 53m 37s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/3/20220802150501/NEC-V-ISCSI : SUCCESS in 2h 33m 10s","accounts_in_message":[],"_revision_number":3},{"id":"7e3715d7719b9564b1e1833bcc654bd828ebf455","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-08-02 09:08:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8dea11164dda4a4587c35dd1aa11b373 : SUCCESS in 39m 00s","accounts_in_message":[],"_revision_number":4},{"id":"9589d5ccea3b9819596ed228d49fa6f9381a84a4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-02 09:24:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/847131/4/iscsi/ : SUCCESS in 1h 19m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/847131/4/smb/ : SUCCESS in 1h 00m 17s","accounts_in_message":[],"_revision_number":4},{"id":"069d7ad14d182f432013e41c9ed2760e42efa605","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-02 09:33:33.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847131 : SUCCESS in 1h 26m 51s","accounts_in_message":[],"_revision_number":4},{"id":"aa92f946b4bb2efd3449a93f1f728d482b1bc9e4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-02 09:39:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/847131/4/check/iscsi-driver-alletra-6k/cf3061b/cf3061b.html : FAILURE in 1h 33m 29s","accounts_in_message":[],"_revision_number":4},{"id":"260de212802b97778fb9cfe851d20f9f86125402","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-02 09:47:58.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/847131/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/86e91b7/ : SUCCESS in 1h 39m 24s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/847131/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/79eaaf0/ : SUCCESS in 1h 28m 36s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/847131/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/bb18a35/ : SUCCESS in 1h 26m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7a30966fe63211dc27a24e14b2d24b47d6c7f085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 10:09:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5c2bbd2108e4ae391e7e249cff8ad08\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be4a29e75fd34c229448dc8793c1fa84 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e3a92bc64c34fc290d4dd887eed2d0d : SUCCESS in 11m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/189bad8150c0405a87571bd0f76f3cac : SUCCESS in 13m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce12e2ce7fc54468a0fc62db9fd82e4a : SUCCESS in 12m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/872d0c8284554ddaba9e2c50557ea1fa : SUCCESS in 11m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/ba9c9284ea0b41bc9f729d442f58c92b : SUCCESS in 54m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/589bbb9e52bb46a78f0964463299f86e : SUCCESS in 1h 13m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/277384e864b74fbebc15ec170dc248b9 : SUCCESS in 44m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/84cc1d25559b48c4b0e8812eb99f5974 : SUCCESS in 13m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a32f5e18658d4f209d283e4ccd6d99ba : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8f31304ed6ce41af98853f23aa99d655 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d64d84bf287b4b9ca1fa46f95423829c : SUCCESS in 6m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d757b99e1a344f6abaa38854b25ef399 : SUCCESS in 7m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/255fbd5b80be4c788b7b70149ede4cdf : SUCCESS in 39m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c51a5eb595244f078f865c898144d745 : FAILURE in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/71b5edd6f3b74202977bc407369af146 : SUCCESS in 1h 08m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8ad23f0b00604fc1a5cb7123ad35e43d : FAILURE in 1h 12m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fd0a9e0c56be4cd8a32e2929f134f820 : SUCCESS in 1h 23m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/009340364908496aa66657b7fe7c73b5 : SUCCESS in 1h 05m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1be5de24b4844e33b05625cf3e1fcef8 : SUCCESS in 1h 15m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e94502e0c63e40c0a3d45cc50b41c0a4 : SUCCESS in 43m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/aaf9e726340249aeb9d23dff3a2a2476 : SUCCESS in 1h 50m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/377ca9d6e75a409fb62c9cb7e8ddc36c : SUCCESS in 1h 54m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4fb82fcb4ee34d1fb15e5fb30082a504 : SUCCESS in 40m 46s","accounts_in_message":[],"_revision_number":4},{"id":"9779c4408fb6a5709dd25895139e94903d8cccc7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-02 10:17:11.000000000","message":"Patch Set 4:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/02_08_2022_10_09_10\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"2f170ad066cb3a2ace72a9dbbd2de2528ac46fdb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-02 10:18:43.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/b7517a66743e4cc2a858ddd36544dcd9 : POST_FAILURE in 1h 39m 22s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/26b6b4f75e764449947dead39c9ae7b7 : POST_FAILURE in 2h 06m 54s","accounts_in_message":[],"_revision_number":4},{"id":"113cbf943513be3c5abd87428b3c43fd3452c380","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-02 10:53:12.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/4/upstream-check/cinder-tempest-plugin-ontap-fcp/01ab5c5/ : RETRY_LIMIT in 10m 27s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/e141f1d/ : SUCCESS in 43m 20s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/4/upstream-check/cinder-tempest-plugin-ontap-nfs/7ed741e/ : SUCCESS in 42m 37s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/4/upstream-check/cinder-tempest-plugin-solidfire-iscsi/6517641/ : RETRY_LIMIT in 9m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"3304616c1526d720985232ce247aff23b9a0d653","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-02 11:09:08.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/31/847131/4/check/DellEMC_PowerFlex/5b52e7e/DellEMC_PowerFlex/1118 : FAILURE in 59m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"970b79fd754d587c5b9d78b5528420ef9fe6ad78","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-02 11:31:11.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1659427473692 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"6b45b7616520163036ba8d8c80863a9be8196724","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-02 11:35:09.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-847131-4/ns5_nfs : SUCCESS in 3h 30m 12s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-847131-4/ns5_iscsi : SUCCESS in 3h 29m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"2b9625725f5277093b24588231a4e14ae22cc625","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-02 12:04:09.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/31/847131/4/check/DellEMC_PowerStore_FC/946dc1c/DellEMC_PowerStore_FC/657 : FAILURE in 14m 40s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/31/847131/4/check/DellEMC_PowerStore_NFS/580c005/DellEMC_PowerStore_NFS/201 : SUCCESS in 1h 25m 38s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/31/847131/4/check/DellEMC_PowerStore_NVMeTCP/e432a46/DellEMC_PowerStore_NVMeTCP/73 : FAILURE in 1h 54m 44s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"1c3eee1e9e6936903d7b14f0a1f559d8539fbf45","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-08-02 12:18:10.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/31/847131/4/check/DellEMC_SC_FC/2c802d9/DellEMC_SC_FC/847 : SUCCESS in 47m 29s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/4/check/DellEMC_SC_iSCSI/cc68080/DellEMC_SC_iSCSI/865 : SUCCESS in 2h 08m 50s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"d3b20ea88c93718c65522cf1e7f6f397e7dd282d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-02 12:19:54.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/31/847131/4/check/fujitsu-eternus-dx-iscsi/06b7442 : FAILURE in 3h 49m 04s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/31/847131/4/check/fujitsu-eternus-dx-fc/e33db44 : FAILURE in 3h 49m 38s","accounts_in_message":[],"_revision_number":4},{"id":"f0c5e2dd262fd2676e5b049db305144d7ba27765","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-08-02 15:35:48.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"a78d8fe903dacae3538b81730d7cd1ead4f63da1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-02 16:27:15.000000000","message":"Patch Set 4:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/352","accounts_in_message":[],"_revision_number":4},{"id":"836544950d639b05f4b1f454d0bab9f0766e7eb5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-02 17:05:16.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/4/check/DellEMC_Unity_FC/f97d6dd/DellEMC_Unity_FC/1332 : FAILURE in 20m 26s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/4/check/DellEMC_Unity_iSCSI/8f3e0d9/DellEMC_Unity_iSCSI/1211 : SUCCESS in 2h 04m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"8405a5cd600542cba3f43157c3fd084226e37ae2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-08-03 02:03:12.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/4/20220803083107/NEC-V-FC : SUCCESS in 2h 00m 08s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/4/20220803083107/NEC-V-ISCSI : SUCCESS in 2h 32m 02s","accounts_in_message":[],"_revision_number":4},{"id":"00e05cca7e8bdebddcea3f81a53703fb4db14b6e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-08 04:31:53.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/4/20220808065911/NEC-ISCSI : SUCCESS in 3h 25m 48s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/4/20220808065911/NEC-FC : SUCCESS in 6h 32m 39s","accounts_in_message":[],"_revision_number":4},{"id":"9a89426382aee25a42408a87476b0973911e8ab5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:34:53.000000000","message":"Patch Set 4:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":4},{"id":"7ade287844efe96648c2cf786e4f32190accc013","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-01-16 14:56:42.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"7f9dce1cbe134c26a45353a377f0c5aba3db9b78","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-16 15:00:34.000000000","message":"Patch Set 5:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":5},{"id":"06b0b0a854d91164059f97a66ac9f72f8093ea48","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-16 15:21:11.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"266f86e217705e1b857698dde39304a039066c6f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-16 15:28:44.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/31/847131/5/check/iscsi-driver-nimble/19ee2f2/19ee2f2.html : FAILURE in 12m 33s","accounts_in_message":[],"_revision_number":5},{"id":"069ae6490c7e789a3c1a4c67eaf830322b6591fe","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-16 16:11:33.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/847131/5/check/iscsi-driver-alletra-6k/223df00/223df00.html : FAILURE in 25m 56s","accounts_in_message":[],"_revision_number":5},{"id":"6b043e55cd76f4854983596f4d87fd835fd527b3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-16 16:34:02.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847131 : FAILURE in 26m 10s","accounts_in_message":[],"_revision_number":5},{"id":"7789ee1eeae34936188e9b8f54c38b940d464639","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-16 16:35:40.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/847131 : FAILURE in 23m 27s","accounts_in_message":[],"_revision_number":5},{"id":"352cb175683160a4f7ec413a4786013d67bf3e6e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-16 17:04:12.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-847131-5/ns5_nfs : FAILURE in 1h 44m 31s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-847131-5/ns5_iscsi : FAILURE in 1h 44m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"90c4d1514f7ecd9b2ea2b6f602eed85f40d4a72f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:11:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4b899737e074eacb035c069a96cdcc2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9f29a1792254097a2332a5feafe7d0a : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7065f45511bd4fc18b58b29a9b5817ce : SUCCESS in 9m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c19b79743af43848dd539b56919b27f : SUCCESS in 12m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4110f7bd9f334389ac6e42a3a7598d12 : SUCCESS in 11m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/4299c25f0294416193d7143883259399 : SUCCESS in 51m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8b6aebccdbe84f8e98fb08df840a6bec : SUCCESS in 55m 51s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e52c3ca3749e4a4da72a59cf3c9aa2a8 : SUCCESS in 1h 23m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57a98e2325c542738b5f7df89706fa29 : SUCCESS in 57m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f017cde7620343d89c0a4d38597583fd : SUCCESS in 22m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4f7216ed145b4afc881bbaace6b64507 : SUCCESS in 5m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c17da06b389546069bd3f61d89bf3fc6 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a9dbed12638d4c61bd85ac67342e5bc7 : SUCCESS in 5m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/01daca35ebf34ce8a99d11edb47edf5d : SUCCESS in 7m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/46a61dcefc484e2d901d004336f77b17 : SUCCESS in 34m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9c32c8aeb5ee4daf81c7aae84f4e82be : FAILURE in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7c0b631901bd40a381f61efc706ee30a : TIMED_OUT in 2h 05m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/03682873c80848a7976ef90372df631e : FAILURE in 1h 23m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ac189dc21e3b441ea52890f56a1e5fc5 : SUCCESS in 1h 37m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/76c2c0559644417b8d9b7164b9274744 : SUCCESS in 54m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0a0b29c63add4a5cb5ec30664dae47d0 : SUCCESS in 1h 40m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1eaa7a4515a14b1ebdb69dbb3be89a76 : SUCCESS in 31m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f51f5b5cce564b099e2f066a7b49f394 : SUCCESS in 1h 20m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6e1ecdcc32de4b4798091c2ce86279eb : SUCCESS in 1h 56m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89ed6de4a9cd4ba29835a67f270b3c7f : SUCCESS in 49m 02s","accounts_in_message":[],"_revision_number":5},{"id":"f579f4a7e1a92d81051d485df9f21d2f8fc950e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:16:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db175324134a4592a37e972a05dd7f40\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/342d96b476184212b3afcefe15bd71c5 : FAILURE in 28m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9356873f5a06406597a037f1948329bf : SUCCESS in 28m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c098aba9afb3527d0c3d2367c06c483b6316c3fa","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-16 17:22:12.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/847131/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/37f1140/ : SUCCESS in 1h 43m 21s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/847131/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d95ce72/ : FAILURE in 2h 19m 27s\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/31/847131/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/5c8d984/ : SUCCESS in 2h 16m 47s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/31/847131/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/884afe8/ : FAILURE in 2h 03m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d075ea00ba30d36cae79aff5fdfefa63a39cc931","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-16 18:12:07.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/039da626179b47d6959ca3ff87e83dc7 : POST_FAILURE in 1h 57m 31s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/ee03bbf234cd4bcc8c5a6712b565d5da : POST_FAILURE in 1h 45m 37s","accounts_in_message":[],"_revision_number":5},{"id":"75647909d2ba976988ef43f1f4d700f8d99b1e91","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-16 20:57:46.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/5/check/DellEMC_Unity_FC/8e881ec/DellEMC_Unity_FC/2156 : SUCCESS in 1h 38m 24s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/5/check/DellEMC_Unity_iSCSI/dd8c0ec/DellEMC_Unity_iSCSI/2025 : ABORTED in 3h 00m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"e37ef3a3897a5f59a6d66cb9f53fef53b05bd528","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-01-17 01:44:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6bf4b2bec6ce4fcda3f6bc72a6244428dfac468a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 02:15:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ab156b164bf421e966ebcae075596dc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/430b2255a30b4f59bf305208100e4fa1 : SUCCESS in 25m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7272bf20a5f4410596240d5c8a67a1b6 : SUCCESS in 27m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9b15de8c673454d4c033939889486ee0ed87541c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-17 02:15:11.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"958e846a7fb1c8bd8ccf5f884ec17e9320c25d14","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-17 03:06:05.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/95e925aa0a6d4114bf4a08ed59b3f213 : RETRY_LIMIT in 8m 47s","accounts_in_message":[],"_revision_number":5},{"id":"9c1932da18b6db49d472a2ff0f10ae7c682a1272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 04:42:23.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8584e83e1dfa42aabab33eae7e654175\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8ed7c42191b4647b81183883ee79ad4 : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f5af37221904475a869f1116998b41e : SUCCESS in 10m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62209773babc4a8d8df1cd3a9338594a : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a995cd76dfbe4378870efc6e273392ac : SUCCESS in 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/09d5e2c7381d4b3c874dbe24d88c579f : SUCCESS in 1h 00m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8180f9b111f240adb89a8c5c851a697f : SUCCESS in 53m 49s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a1142bb3fd10488b8628cbb6ef3060ee : FAILURE in 1h 48m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d54cda8673c43ad97cb698fd14ab8da : SUCCESS in 1h 04m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e6519d46eef34f258d92d5f278c7b28a : SUCCESS in 16m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b3873a7f8893496680274bb03083b8c0 : SUCCESS in 6m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8538663820f1493a91206524280590f7 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eca0e36f3ac84e059182c51568f6f1ff : SUCCESS in 5m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/83c55b82dfb84d459144c6e59ddfefd0 : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/315d4565a4a24d4183a62e762ec8e087 : SUCCESS in 52m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/62cbcec712304a16be73f303dfbafd0f : FAILURE in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/96492b9309184ef08eb75182f83b231e : SUCCESS in 1h 11m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/89c6a14943a34a4c996dbe7a185e167e : FAILURE in 1h 31m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d91014127779468091bde92526ec8068 : SUCCESS in 1h 39m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/15f26600493a4c9c99addec53cf7ada9 : SUCCESS in 1h 11m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/61732565d2524dda8307b957ba178e0d : SUCCESS in 1h 28m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bdfc90ebf2ed40159294f22c8857628e : SUCCESS in 47m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8de6d62a366c47168e611c355075c3b4 : SUCCESS in 1h 43m 09s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/26c083a21e0f4e95aa7677de932303cc : SUCCESS in 2h 32m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/189a573a455348d58af4b291b0a62cf8 : SUCCESS in 51m 44s","accounts_in_message":[],"_revision_number":5},{"id":"4eb59ec057eed963266d194aa69fe0d9b397e79d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-17 07:22:08.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/31f7c0da38184c3fa5476db43362c59f : RETRY_LIMIT in 8m 52s","accounts_in_message":[],"_revision_number":5},{"id":"73560cce897f8a57daeced41dbfbe84369bb3112","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-17 07:59:35.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/847131/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/76058b6/ : SUCCESS in 1h 56m 54s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/847131/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/16c65d7/ : FAILURE in 2h 27m 54s\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/31/847131/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/79d2a6b/ : SUCCESS in 1h 47m 18s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/31/847131/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/c2e6210/ : SUCCESS in 1h 54m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7e2b384a61e77652ee390909f5760e21d1340a11","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-17 13:03:03.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/2a342700b5f0458585e1787d0d1e7e6d : POST_FAILURE in 1h 57m 03s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/9437cd61812840e4b6d2c379b6ac2d13 : POST_FAILURE in 1h 25m 54s","accounts_in_message":[],"_revision_number":5},{"id":"dc851b51f555f34c00f87df79ce97d3b4c07c7a7","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-17 14:22:30.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/847131/5/ : FAILURE in 48m 30s","accounts_in_message":[],"_revision_number":5},{"id":"590b75cde356db53a5b33e3cced6801e97f7ec42","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-01-17 23:39:39.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/5/20230118024024/NEC-V-ISCSI : SUCCESS in 2h 24m 06s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/5/20230118024024/NEC-V-FC : SUCCESS in 5h 59m 12s","accounts_in_message":[],"_revision_number":5},{"id":"eae712545d76e34536da1685dee1af6facba04d4","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-01-20 23:34:56.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/5/20230121014012/NEC-FC : SUCCESS in 3h 29m 22s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/5/20230121014012/NEC-ISCSI : SUCCESS in 6h 54m 42s","accounts_in_message":[],"_revision_number":5},{"id":"bd8bb7f759d38440bffb3049e6a0e57302c36f5a","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-01-24 10:28:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9d218d0166ecc017ede28c19f8e4fb3f1fd1450f","tag":"autogenerated:zuul:check","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-24 10:31:28.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"bf8233bdda09ec6ca236ee18952b21bfff555362","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-01-24 10:48:19.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/205acf2535f7466fb5fe2c1717550fed : POST_FAILURE in 18m 29s","accounts_in_message":[],"_revision_number":5},{"id":"d2592e9dcd323e755a6fba316c227cd41c6bb07f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 11:09:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dc3c0d55bba47ff9500afb836a9def5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d215ab88aa8e4984a47d66a7d41c5e0e : FAILURE in 36m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/86bf19fef2024a57a5c76437efff2c00 : FAILURE in 37m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a22253cb1252672b229755f69c90782f4ee08e05","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-24 11:47:13.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8950465673e241798d6d024349611566 : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":5},{"id":"73fd1d16e3599ffdc72d0ee27b2346433f4d0bd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 12:45:09.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c592d9b2d3b64cf1bec511e5b984aa4e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e755e31deef46cca4739e2ed86e0149 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81bfbafbf7f9459c8ec1a7efef481769 : SUCCESS in 10m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0671b7f5f9b9450d98c0a633935cb1bb : SUCCESS in 12m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc8b48dc0c2045f2bf727c8f9e474af5 : SUCCESS in 11m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/8dbd6ad75c4d4ad1b78a7b8cad26967c : SUCCESS in 55m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2b97253dfd394eb68e017816d0bedf10 : SUCCESS in 1h 03m 18s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fc9b15568b0341169458ca59c09cd41b : SUCCESS in 1h 18m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf029640c5c64ca2931131da2270b2b5 : FAILURE in 59m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/802371d32de44d09b55695d9a0443565 : SUCCESS in 15m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dc53da25d6a149dbb2a191b38afb9b0c : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b1dba39458c14cee8dde8c3650269901 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/68f07e55cc14464992e3db89b8439ce6 : SUCCESS in 6m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a8efe51d80f24041a33253ffbee3402c : SUCCESS in 5m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4a2b8e9ad850406598d13388b454a84d : SUCCESS in 45m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7ad5228dcf824c4cb4f1c049dc319bbc : FAILURE in 5m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/625c7b5fdbdd483ab733a3a0e70bba20 : SUCCESS in 1h 19m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e2cf4849c80d4d89a7bfc5333706cdd5 : FAILURE in 1h 16m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fc85f40c7261418195caa4f60500e375 : SUCCESS in 1h 57m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9b0e03cc26984155940ae689dce4df59 : SUCCESS in 1h 15m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9c0e682d40e44a8f8df559b7b7f991d2 : SUCCESS in 1h 27m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1632ebe1e5904da0ba414f7d1f4be098 : SUCCESS in 44m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/140b6ee79fa94e15a7786e44c8d535d8 : SUCCESS in 1h 31m 21s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/145cdfb0d4d24ea5a9c47969bb17dab4 : SUCCESS in 2h 13m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/770f59ce5d90463d94722270e37ab749 : SUCCESS in 52m 00s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfb6e6c0e59341059e21508e303a1895 : SUCCESS in 1h 11m 28s","accounts_in_message":[],"_revision_number":5},{"id":"76c2a9ef781b629f28c528e319740a1e2b440a04","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-24 12:51:58.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/847131/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a500fdc/ : SUCCESS in 1h 30m 36s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/847131/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3e703c0/ : FAILURE in 2h 21m 32s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/847131/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/3745536/ : FAILURE in 2h 10m 10s","accounts_in_message":[],"_revision_number":5},{"id":"fe46f9c74460f55ddb47a53b9ee628c362db8bbd","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-01-24 16:05:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a386924f32966e52f3be171891ec915d3aa1de74","tag":"autogenerated:zuul:check","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-24 16:07:29.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"f3855d7f0830aae9a011110fe7a2e653fa6e250e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 16:47:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c8ca0133d90448ea9f0ea30d6ba0972\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7db9f7fae2d2487a968d34712ccb2ab4 : FAILURE in 30m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ca7b2ec155f54bdba43d4ea83fb325bc : FAILURE in 27m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"618ea7be710d631ab71fc93879237eb28f57dd39","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-24 17:18:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/495b69af6b3b4c04bee2c4ed838fff7c : SUCCESS in 1h 11m 31s","accounts_in_message":[],"_revision_number":5},{"id":"b084ad446faeddd5aef843b5cf5db48aad5fedac","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-01-24 17:39:46.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/d0125967476d4cb2814aed2e39a6fb3e : POST_FAILURE in 10m 46s","accounts_in_message":[],"_revision_number":5},{"id":"c46e0b75c12f001eb7f5abcf28e33f0280c43817","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 18:27:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2afa5f604ae243a9a7d55f159858cdc0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7a09c57f3d4480f8e539dc9746aaf36 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/773786fefa93404ca834e29df66a335e : SUCCESS in 11m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e56d6d585a2d4459801bf0a63d45f37b : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1088e7032bb04995a6b1a7a2324f5d91 : SUCCESS in 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/21f3866dc05b4f42beaeae87358fc147 : SUCCESS in 57m 01s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/008dd62967164a429e9a4529edd93aeb : SUCCESS in 55m 10s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/35227f14d14a4346a05cbc4fb86bc89f : SUCCESS in 1h 15m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f2423f5eb904672bd3563ad0f691b1d : SUCCESS in 51m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/65dcba9fd38b4f16b33ceb9915058716 : SUCCESS in 14m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7f8d975adbac4c6b9f69468d66e2613b : SUCCESS in 5m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/52fd4e84e8854764a39f6fc29db403b5 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1ae4742ca75f4cfd8ff68ef7d542c78d : SUCCESS in 5m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/85f2cafd91cf46b89a27d236d6ccd704 : SUCCESS in 7m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1313b6faeca546998681be15578a5678 : SUCCESS in 48m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4f976272e82740cd9c8c3729dbc4df27 : FAILURE in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a33b61c180914fcda446ee12d24dc097 : SUCCESS in 1h 16m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/be546e58a5794363afb2f381f4d31324 : FAILURE in 1h 25m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/16b0d3cd03a74ecea1ee3a8e9a15500d : SUCCESS in 1h 57m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/343903ec34e24f9d84da41fb9619e36c : SUCCESS in 1h 03m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/17ca5388b06d4d259f5c2f36a1dae003 : SUCCESS in 1h 51m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fcdf18365f304ad3bfc8df771bbf4a40 : SUCCESS in 43m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/51a23db8f5fe472c86fca642a458385d : SUCCESS in 1h 26m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4a41c1c9465b4b2c9c36c96b01cf499d : SUCCESS in 2h 06m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d6f94c0e0f04a99a59719584fd2c180 : SUCCESS in 46m 31s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/455db3ff609d455ebb2fba9e11fee410 : SUCCESS in 1h 15m 21s","accounts_in_message":[],"_revision_number":5},{"id":"b31c7938f73cb34109d2f8f076b9c5653ef1f6b0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-24 18:35:37.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_01_2023_18_27_13\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"eff1f64b125bef7c92423aa2742038b7fe85e46e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-24 18:50:27.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/a484bca321914cfe9f580e84567cce93 : POST_FAILURE in 1h 39m 25s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/774197ccd01e463893a107368fa849dc : POST_FAILURE in 2h 21m 41s","accounts_in_message":[],"_revision_number":5},{"id":"27beb66ff981dc822797c4b090c562e09946e24d","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-01-24 19:05:05.000000000","message":"Patch Set 5:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/31/847131/5/fc : FAILURE in h m 0s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/31/847131/5/iscsi : FAILURE in h m 0s","accounts_in_message":[],"_revision_number":5},{"id":"cfaa5c1ef81ed51cb1a046f27f202544c72bdcf6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-24 19:28:55.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/5/upstream-check/cinder-tempest-plugin-ontap-fcp/72ed10a/ : SUCCESS in 1h 01m 01s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/636bd08/ : SUCCESS in 42m 55s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/5/upstream-check/cinder-tempest-plugin-ontap-nfs/7ae2e9d/ : SUCCESS in 41m 42s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/9c74d1a/ : FAILURE in 31m 28s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"b00afc5907fb318a1763dc0a8233200919c75f42","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-24 20:44:15.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/847131/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b2d8c65/ : SUCCESS in 1h 50m 11s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/847131/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/b7ecbaa/ : FAILURE in 1h 56m 42s","accounts_in_message":[],"_revision_number":5},{"id":"a6bff291394be2be20d87b9b16ea7656dbb55fa3","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-01-26 02:24:16.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/847131-5/HPMSA-iSCSI/ : SUCCESS in 1:36:05\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/847131-5/HPMSA-FC/ : SUCCESS in 1:16:37","accounts_in_message":[],"_revision_number":5},{"id":"78944c648a8ff16c8393967732a4ba78b0fc04e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-01-26 12:35:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"48926e1699a4f589ff3642417e37a69aa0f157a3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-01-26 12:35:19.000000000","message":"Patch Set 6:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":6},{"id":"9f7868ed28bd11b9f03e6fd15c5932c4b7942329","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-26 12:36:24.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"6968cf0a1725c79b35126fcbd732995ab3cf385c","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-01-26 12:44:41.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/8d4071e7023443a381fc91b9c7b92645 : POST_FAILURE in 6m 22s","accounts_in_message":[],"_revision_number":6},{"id":"4141fcd5ca7b739fcea28716284fd5304850e1cf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-26 12:44:51.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"d12f3654e43a8028b1582fa7b492e0e5ce81cdca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 12:50:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46035d4b0b794f548f90a72a418344d0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a82c5214fd84b34b9b3e4606c4e8ed8 : SUCCESS in 12m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f3f23359338a4a2c9d6a839d3644cfd7 : SUCCESS in 12m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"aa146939c3c00ea0a1e0a2209a7aa12164c70643","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-26 13:07:54.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/847131 : FAILURE in 23m 51s","accounts_in_message":[],"_revision_number":6},{"id":"fa9b70741bb1c59a0dc3b58a8dfa346cda6f4f74","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-26 13:16:32.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/847131/6/check/iscsi-driver-alletra-6k/81368e8/81368e8.html : FAILURE in 25m 46s","accounts_in_message":[],"_revision_number":6},{"id":"b3eb15250febd9ac674b041aa4d60dfc52587e09","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-26 13:19:31.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847131 : FAILURE in 26m 36s","accounts_in_message":[],"_revision_number":6},{"id":"a51947761de903cb7876ba109973389d3b03b03c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-26 13:48:00.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d94526ee6d444f3790dbe30037894593 : SUCCESS in 1h 09m 49s","accounts_in_message":[],"_revision_number":6},{"id":"2b5018c40c66f973ff2f94f8ce0be9b01987a578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 14:46:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8878b48f71a54c8791d4e252e2cd8c63\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd982b8775742c693ebf49621efb4a3 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3db034249264e5dbc816586ee506a3b : SUCCESS in 10m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1dfca0183aa4872aecaccdf2a74eae9 : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db87838b20140bb98267e288a3d494c : SUCCESS in 11m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/beb9e65fe13346e6ae778e93685151e8 : SUCCESS in 1h 05m 12s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/aaf83ece69c64d729b9428152e958bff : SUCCESS in 47m 18s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/916b4ac5f3704dc6ba68593704db4b35 : FAILURE in 1h 17m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45f7c1c1e2f04a04a3149a615f866ec1 : SUCCESS in 55m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3e9c0c9450f4443095cc50e9214eb976 : SUCCESS in 16m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/20a3bd7ce29e40a0857e4d3e9faca14d : SUCCESS in 4m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fe1713b104624845bfbb70beafa3d106 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f602c59322a54cfe877f58e57633039b : SUCCESS in 5m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ebf0f9e2e2784540b361ccfdd09658f7 : SUCCESS in 5m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/03d830c3c9dc4dd3a0378d8f88042aea : SUCCESS in 35m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d45db8121748436a9b53533e6a8470fd : FAILURE in 5m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a3850b4d510a49b2b1e4f93236cca0ab : SUCCESS in 1h 17m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0ce4d3b122614b83956c8994eb7639b1 : FAILURE in 1h 26m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/14e11ee069614ebf8f0822ec34ddd778 : SUCCESS in 1h 43m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/566d6749b8834dc58e338bd838556f92 : SUCCESS in 1h 02m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3a5b9da62487477db9ccfe7e728ff853 : SUCCESS in 1h 43m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/951f2c61f7b04712bbc2b46537d87fdf : SUCCESS in 40m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4305ccc7cc7e4b6894dc44162d76c784 : SUCCESS in 1h 39m 47s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3c4cdf6b504c4ea2b06291daf9ad0989 : SUCCESS in 1h 56m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1af3637954c841668ea292aa6a99aad5 : SUCCESS in 54m 27s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b282a4c0ab1473e95aaa745c9daf3f6 : SUCCESS in 1h 18m 22s","accounts_in_message":[],"_revision_number":6},{"id":"1ab03c70f732d9e53e4abb89972c8a3e37962b8a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-26 15:12:02.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/847131/6/ : FAILURE in 1h 15m 24s","accounts_in_message":[],"_revision_number":6},{"id":"ce2ae3e1cc832724382186d5b249af41556120f9","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-01-26 16:18:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"af45a4fdeaeffdfe972797b47dd6e0c31186c999","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-26 16:19:02.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"9580bce3fd1dd4dbe572ab2c50c919c9b5eefd9a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-26 16:22:49.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"7aeadc60daced83a1eb2db3d6947ea85a9ea7f63","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-01-26 16:23:59.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/3cefd2343132439fb57575831c2cf4c8 : POST_FAILURE in 5m 13s","accounts_in_message":[],"_revision_number":6},{"id":"ec7951dce4545f907141608b988cf9fe0077948c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 16:27:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bec51c2fb36942b7ac4f2e1c9254bb39\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a618f855e9ab487ea9ad32c642585d22 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/663b0fb95ed14e19a57282e719e178ee : SUCCESS in 8m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"41961881b145b9b323d620f749084e7c8362462a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-26 17:31:34.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b494c40bfe754e8e8ede81d0ac6abc44 : SUCCESS in 1h 11m 36s","accounts_in_message":[],"_revision_number":6},{"id":"8a4c3ed703d37a757a59777b6bdb87b040754259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 18:21:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83be7314028d4055acbda8e452587864\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a51ebe583fe49abaed9ff1eb0e1de81 : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48c9a2cc95834b4ea7427d480289d8c5 : SUCCESS in 11m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b09b36ed051c44c28197f8cabbea6a8b : SUCCESS in 14m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efec20c81cb9465aa60b36bf682718b2 : SUCCESS in 12m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/44aec9bb53124cfebe8bcb583acd53f5 : SUCCESS in 52m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d2b2ce41f10b466ebfede2f0381f35b0 : SUCCESS in 1h 03m 53s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/67ef53e534904165ac15fd4a7bfa32aa : SUCCESS in 1h 41m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c179ef40d72147a2a7c474d6977eefbf : SUCCESS in 49m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/04f7c56f8ef94fb896c88f439eab4c69 : SUCCESS in 16m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/06d338920fb84fe59c1e13d654ed9a3b : SUCCESS in 5m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/130a70ca433c447f828c33e1c459d34b : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2d4b8a6376ba48568956641fcd7b1749 : SUCCESS in 6m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ff5508e763ff47adbbe991c5c4376414 : SUCCESS in 5m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/16e2889a790d494f96496127ae8ed9ff : SUCCESS in 37m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/065bd179a247452b85a54928d3186fea : FAILURE in 6m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fbf808effee848d389db2cd9fdfe2c8f : SUCCESS in 1h 32m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/050bc80cec7044db8be18e3271b0ae42 : FAILURE in 1h 21m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/33edea544bed4da59425f0f696e30037 : SUCCESS in 1h 37m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6224e33374d04a97a7bbca1e69e4ef51 : SUCCESS in 53m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/91bb447077ec4d6084786d13875b6155 : FAILURE in 1h 49m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8c43384981ab46ee9de9faa18e59e906 : SUCCESS in 47m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3d73ec0fa843439da790e3787c784731 : SUCCESS in 1h 25m 27s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5dd6bb2f157247699209d284071812fa : SUCCESS in 1h 52m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f27b639ac04a4928b5107c16ed1d1160 : SUCCESS in 43m 52s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98be14dba1f94f0296cf8135ceecf200 : SUCCESS in 1h 09m 35s","accounts_in_message":[],"_revision_number":6},{"id":"7e0cf6e6306c7c51da21e1062978ad4245af0564","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-26 18:29:53.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/26_01_2023_18_22_08\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"f63d5b0e25ffbb8f783f3e02459dfd2e6a96c659","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-01-26 19:05:03.000000000","message":"Patch Set 6:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/31/847131/6/fc : FAILURE in h m 0s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/31/847131/6/iscsi : FAILURE in h m 0s","accounts_in_message":[],"_revision_number":6},{"id":"0d855ade283ce3022b36870731070854f2f96843","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-26 19:24:15.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/6/upstream-check/cinder-tempest-plugin-ontap-fcp/8762a6b/ : SUCCESS in 1h 01m 36s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/f73e694/ : SUCCESS in 43m 05s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/6/upstream-check/cinder-tempest-plugin-ontap-nfs/aa50cea/ : SUCCESS in 42m 48s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/f1925ad/ : FAILURE in 30m 05s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"fb49ac22210137fa97769b41c45008cab5986a91","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-01-26 20:41:02.000000000","message":"Patch Set 6:\n\nUNSTABLE build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1003","accounts_in_message":[],"_revision_number":6},{"id":"5b525f73410ddce82e7aa0de7e595fe41abfffda","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-26 22:59:28.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-847131-6/ns5_nfs : FAILURE in 28m 2s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-847131-6/ns5_iscsi : FAILURE in 10h 23m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"62e7133bdc6fb998560e1bd4d730e84f03630370","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-01-27 10:46:06.000000000","message":"Patch Set 6:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1007","accounts_in_message":[],"_revision_number":6},{"id":"e3a6c3afbc49b2841c8c687f0209ec7848c6865d","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-01-27 21:02:06.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/847131-5/Seagate-iSCSI/ : SUCCESS in 1:55:23\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/847131-5/Seagate-FC/ : SUCCESS in 1:33:18","accounts_in_message":[],"_revision_number":5},{"id":"fc7a5d81a33a4b004020babfbe439b001e8bdcda","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-01-27 21:04:57.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/847131-5/PVME-iSCSI/ : FAILURE in 1:55:40\n- PVME-FC https://os-logs.seagate.com/dell-ci/847131-5/PVME-FC/ : SUCCESS in 1:33:32","accounts_in_message":[],"_revision_number":5},{"id":"e83c9f271ca997147ec5479abf172c79944ae944","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-01-28 06:31:35.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/847131-6/HPMSA-iSCSI/ : SUCCESS in 1:36:22\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/847131-6/HPMSA-FC/ : SUCCESS in 1:16:42","accounts_in_message":[],"_revision_number":6},{"id":"3139231df9c33ed2323ea1ef8bac04248a881219","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-01-30 05:50:52.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/847131-6/Seagate-iSCSI/ : SUCCESS in 1:54:43\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/847131-6/Seagate-FC/ : SUCCESS in 1:32:35","accounts_in_message":[],"_revision_number":6},{"id":"fd1742c232abdb51b4997e7e7d7c8639d79c8e15","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-01-30 05:57:51.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/847131-6/PVME-iSCSI/ : SUCCESS in 1:58:22\n- PVME-FC https://os-logs.seagate.com/dell-ci/847131-6/PVME-FC/ : SUCCESS in 1:32:03","accounts_in_message":[],"_revision_number":6},{"id":"6de41c28d04e530ddcc3f1982b002b88a0ab4484","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-02-08 13:33:35.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"0f16a49f052372db451bf0be9c957cfd9a5eedbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 13:46:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebcdf1197347406cb70febcf34afea75\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4745fe393f7c41198a7262bbd8a80548 : SUCCESS in 11m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aa4a10cf5fa040f5ba69f270a8adf2a6 : SUCCESS in 10m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"65fa09c95ed9b99946f687d55536d301a635a560","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2023-02-08 13:54:12.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/31/847131/7/check/dsvm-tempest-zadara-driver/fa9d90a : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 20m 16s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5222e3e3f78e60b9ef84acf207f5aeb5840550ea","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-08 14:05:01.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/893511a9bad44caa9e60ef668a3068e3 : POST_FAILURE in 28m 23s","accounts_in_message":[],"_revision_number":7},{"id":"9b764e63461d8b0c1f595fa292ba148485745765","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-08 14:08:16.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/847131 : FAILURE in 23m 25s","accounts_in_message":[],"_revision_number":7},{"id":"f52a7e2b98da979e99216d0e34abcc249e6c4357","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-08 14:09:00.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/847131/7/check/iscsi-driver-alletra-6k/83e0044/83e0044.html : FAILURE in 27m 00s","accounts_in_message":[],"_revision_number":7},{"id":"8353c1b7fc95659a969a3c771c8e497c0d34966b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-08 14:11:37.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847131 : FAILURE in 25m 59s","accounts_in_message":[],"_revision_number":7},{"id":"a775a2e8e63b25eb95eb851843a18312dadfd674","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-08 15:04:21.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/be349a2e29d045589970b2cbc61961eb : SUCCESS in 1h 27m 36s","accounts_in_message":[],"_revision_number":7},{"id":"570080db90b50604b6e90891c754e2fde8ec77b6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-08 16:07:36.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/847131/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d4dd327/ : SUCCESS in 1h 43m 53s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/847131/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/cf5cab2/ : SUCCESS in 1h 46m 57s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/847131/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/80e84e5/ : SUCCESS in 1h 33m 11s","accounts_in_message":[],"_revision_number":7},{"id":"42b51572d70d587b9a436fb48eb3cc895bb8fd27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 16:33:52.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cff67480c2944c49539a2a4e50ef9c4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49f7b875b55c4ccda9eaf3e466ea0ef3 : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/428ae3d5f62649469d95f8efd127fcb4 : SUCCESS in 11m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8b9ccea886349ef93bbd2189c257821 : SUCCESS in 11m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a73c464ee514e3e979c5106fd79ad52 : FAILURE in 6m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/fc487a2748f74f338634d2e6d7fd7d88 : SUCCESS in 52m 51s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d165b3a315e540299b9ef1c7d83b1502 : SUCCESS in 1h 16m 48s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bb6629a3f7c74971ac96efdb8f926455 : SUCCESS in 1h 33m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2455c3ce503d4230a0dc53dd84a92e6a : SUCCESS in 1h 10m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/929fea351ac44ed6b2ffef25433544b2 : SUCCESS in 14m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/286d5f30444e4f4f9e5d39fb9a13cf6a : SUCCESS in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5f7668be740b449083e5aea8f66876c6 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e790f382b20c4457b759c30e0bcb508c : SUCCESS in 6m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8fb504f16b874da8b43a7d513939c492 : SUCCESS in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eacd45f6d5a94a0aa5dccc00869446e6 : SUCCESS in 35m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ba50869f38f94d259bd484efee756b4d : FAILURE in 5m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/196feb8410cf427ab9a3945f032223ba : SUCCESS in 1h 17m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ee69f612673b4ef29c79dc23bb05473f : FAILURE in 1h 30m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4886c308560a436cab1dc44f43144894 : SUCCESS in 1h 57m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/562d927e98e34ad9bdd5edadaf4da33f : SUCCESS in 1h 14m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7e57262a7dba4cedbec6f3e6541e8e34 : SUCCESS in 1h 33m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/283307eadd4e4c6e9869a6e10f4c95da : SUCCESS in 42m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ce93e527aaa144d9b461e03ce6cb7754 : SUCCESS in 2h 00m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e2f216677a25442d87bf8f3049e6180b : SUCCESS in 2h 16m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e00d2e831ee84112b6102e08be75350e : SUCCESS in 40m 59s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb3c7cc308ce4b92bbeb7203cf38a15a : SUCCESS in 1h 08m 47s","accounts_in_message":[],"_revision_number":7},{"id":"092eb77066b3bcfa2ab5c34980d56a3eaa01e89e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-09 00:43:40.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-847131-7/ns5_nfs : FAILURE in 11h 9m 33s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-847131-7/ns5_iscsi : FAILURE in 53m 4s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"502c7b04fbf6744d68b2b7cc05439a0b1da07e1c","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-09 01:00:39.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/847131/7/ : FAILURE in 57m 57s","accounts_in_message":[],"_revision_number":7},{"id":"6f4d1a785c08a93b9f90ab699ed0d3aa3a25d4b8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-09 01:07:56.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/b1892011dc6f4262b6460ac9950b3f8e : SUCCESS in 1h 58m 13s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/57ce07869e3c49468c5fef64b1ff2b08 : FAILURE in 1h 43m 00s","accounts_in_message":[],"_revision_number":7},{"id":"ddb31a59c8a7614d476928ff5c07c364b07aa151","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-02-13 09:20:32.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8d4ae6b034e04ff1c1f43fbec70377c526aa745f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-13 09:47:24.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/847131 : FAILURE in 23m 31s","accounts_in_message":[],"_revision_number":8},{"id":"e71da080e3ec1ac148230ae01c91a301d13372c6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-13 09:48:47.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/847131/8/check/iscsi-driver-alletra-6k/375fdeb/375fdeb.html : FAILURE in 26m 03s","accounts_in_message":[],"_revision_number":8},{"id":"12c77a0cc9569fd5c9930341a60d458b73138135","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-13 09:49:37.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847131 : FAILURE in 26m 14s","accounts_in_message":[],"_revision_number":8},{"id":"4eb1ca70ea8e49373a89380c1e7ec7f5d9114b2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 10:01:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f57938136850404993964cab3e1df0dc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aab2d18b876140de8bb5035e8ec57f4a : FAILURE in 34m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/16c8ebea3cda4055bc85464e846ec2c4 : FAILURE in 37m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c36a30058d3df4a075e152fe7780781757b58ec0","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-02-13 10:25:45.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6ee5ab16ee5f8c22ebe724650b3ce57e5fbd16ba","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-13 10:26:07.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/847131/8/ : FAILURE in 54m 55s","accounts_in_message":[],"_revision_number":8},{"id":"717bfeb9efb827676d1140a47f49ddca04af574c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-13 10:26:07.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"8d9e1b1bfd7ce2004fa42cfa62daa03ab20a6dcb","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-13 10:29:06.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"9e15b1c65039552f3ae75e55985448d2a0e9510a","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2023-02-13 10:48:07.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/31/847131/8/check/dsvm-tempest-zadara-driver/e6bfa3d : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 20m 26s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b5048b89d09668f9cb01936723965dbb01ded0e9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-13 11:19:02.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/eb5381acff544395a96cec6012fc2241 : SUCCESS in 1h 17m 03s","accounts_in_message":[],"_revision_number":8},{"id":"0f979e0d343b82a5c182350c7a562fcf3314b32c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 11:32:46.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78203f8c158b45a7abe5631a42d3add9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/969e47901ae54b0f9d1892d1a577e39b : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/278773de2bea461dbc15071fd76a2900 : SUCCESS in 11m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f3c3955547d4de39fcff5c63ddceae6 : SUCCESS in 10m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b996f0369a974b53949ee2f7dad0ace8 : SUCCESS in 21m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/b7c6f6c0e4894a88a30b9e66d6a0573d : SUCCESS in 45m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/867b611e5e8649ecac68120ce753130c : SUCCESS in 1h 02m 59s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e19a94be55d34a80b504af79e3ec89db : SUCCESS in 1h 03m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9a16e3a38664c1f9530dad8c4bf018e : SUCCESS in 43m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/843bcc681b6a435b8706536faee16451 : SUCCESS in 15m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/47093fde696a4248bfbe0f1e01353e2a : SUCCESS in 4m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/56f1cc2d087e402aa1a09d4451715953 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5b7cf73f97c04363859ec11481237ac1 : SUCCESS in 5m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8000e72a43e34012b3e35ba716729810 : SUCCESS in 5m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bf3125c38c9340f9a7f83d06d571eec0 : SUCCESS in 33m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db36ecdd69214ee7ad23c7fc66b2eef5 : FAILURE in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0bbb33f20f8b4ddfada1c3f0cbe0e079 : SUCCESS in 1h 00m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0b15cc0d4d8544c3a3c15a0f74fc0e43 : FAILURE in 1h 17m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7c5e0d6e5a654e08a1dc12e71f0145fa : SUCCESS in 1h 47m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/541d6f4fb6cf4553a75d97539e731d33 : SUCCESS in 1h 17m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a0a4ff16c5a246b38382435e43fc3291 : SUCCESS in 1h 12m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c1b18a6630a442dfb77bd4ecf99b875a : SUCCESS in 29m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3947b6d3d118422591ea63f2cefcbcc6 : SUCCESS in 1h 42m 57s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1de34540c6964c38a2b5dc8bfd54487f : SUCCESS in 1h 46m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47fd092f4e734c82aa688b05bcf7b942 : SUCCESS in 53m 47s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89dc2cb3c7c341af8d5618d72f58eabe : TIMED_OUT in 2h 04m 13s","accounts_in_message":[],"_revision_number":8},{"id":"8ad7c9f44dfcfc8ce3f33d3941404355a04d2dbe","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-13 12:13:59.000000000","message":"Patch Set 5:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/31/847131/5/check/fujitsu-eternus-dx-iscsi/1de32fb : SUCCESS in 1h 43m 15s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/31/847131/5/check/fujitsu-eternus-dx-fc/de583e8 : SUCCESS in 1h 34m 00s","accounts_in_message":[],"_revision_number":5},{"id":"1c01bf90356fa224e104f0276296bd085d072842","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-13 13:24:20.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/847131/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/ff7abba/ : SUCCESS in 1h 55m 58s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/847131/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f7207fe/ : FAILURE in 2h 18m 33s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/847131/8/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/4cb4b97/ : SUCCESS in 1h 46m 01s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/31/847131/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/d7f8039/ : SUCCESS in 1h 54m 03s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy pure-devstack-cinder-tempest-nvme-roce-aio-jammy : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7c12d98d5361ffdd96774380c986536f641b824b","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-13 15:08:05.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/b1250a667a8e44f59e7db8969c34e8ab : TIMED_OUT in 2h 01m 08s","accounts_in_message":[],"_revision_number":8},{"id":"7c062bbeda13d3dc7b5446176528633f76782c90","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-13 16:58:23.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/8/check/DellEMC_Unity_FC/a134433/DellEMC_Unity_FC/2257 : SUCCESS in 1h 16m 37s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/8/check/DellEMC_Unity_iSCSI/c7573c7/DellEMC_Unity_iSCSI/2117 : SUCCESS in 1h 55m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"d714ef47f464a7801c168ac874d8d7b45ff2f926","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-02-13 23:51:53.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e10f704f8a3abfeee521c0894d4675b32316f90d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 00:06:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6307c55bd19b495d8d9e43c32ddf1524\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/53d2a6a2b3244901b9339d843fc28f70 : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5c4b23ec47ca4218b53f0d18d52d21a4 : SUCCESS in 13m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8c87cbbf8969ced0282782008622b7c921e88e04","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-14 00:14:13.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-847131-8/ns5_nfs : FAILURE in 52m 29s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-847131-8/ns5_iscsi : FAILURE in 14h 2m 52s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"57ae174b0d229e7d1dfbec91d5c1b27dcc3d498d","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2023-02-14 00:48:14.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/31/847131/8/check/dsvm-tempest-zadara-driver/b76b1da : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 56m 07s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2a7c2fc59cc14b1080a6329a9485658f9283e8c1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-14 01:15:33.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/dc4e465a06d6465f98c3a364e910f322 : SUCCESS in 1h 21m 40s","accounts_in_message":[],"_revision_number":8},{"id":"2272c721b9548acec985a1e02d468f7a39b5cd3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 02:16:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0b653ba655942c2bf058029fe57b6ee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0743da2c5ac941c7826896d3e0d6fecc : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a6ec5f94a664b9082b87491fa8d1af7 : SUCCESS in 9m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/296d5176297f4b3e8f597bf02a03ef93 : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba552bf401d24005bd638d8ccdbfe884 : SUCCESS in 13m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/785e64bfb3894056bec65461020c32eb : SUCCESS in 56m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c6dcb10151e94d3abbb55248bfe0c547 : SUCCESS in 1h 16m 23s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6a85b23d6aeb45c7b55b1e215aa04b83 : SUCCESS in 1h 27m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05f7e437ef1a4479b4327f520fd6b347 : SUCCESS in 39m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/404d4b9679584bf1aa149aa6c07ab676 : SUCCESS in 16m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6e462153dabf4063857f74ca4b1eaaa3 : SUCCESS in 5m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/87edda1fb0924822a4d4f0f2cc6c9a76 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/45927d507338432a850ddac3ea3e31ea : SUCCESS in 5m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9df233c8d4514b60883960a2ee01573a : SUCCESS in 5m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a04d7ae657f04eab81bee183295b951c : SUCCESS in 43m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/068cf65cdad046ffbfdccbda4becf31c : FAILURE in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a9741d087da74bc2992cdc5cec51b5dc : SUCCESS in 56m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b69cc9862ca74557a8d1e79a70f65caf : FAILURE in 1h 28m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0b27c838334f4b4fa647c05bc01cee4d : SUCCESS in 2h 07m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a05be9b41d3b4b32bf1b55e64cce001b : SUCCESS in 1h 01m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b907d11db0954f22a2912fad0d827ab5 : SUCCESS in 1h 48m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1779bc054c404f18b6bde92a1ae4200b : SUCCESS in 36m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/79ecbb70405f453aaa8708daed2af3df : SUCCESS in 1h 25m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e13f819a7cd1488987e9f65e7f20eed4 : SUCCESS in 2h 18m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40be3ef6fdb74c89917fe5e5c1a59c12 : SUCCESS in 1h 06m 20s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a25e7d0ae9604557921ba559e6a52209 : SUCCESS in 1h 23m 13s","accounts_in_message":[],"_revision_number":8},{"id":"670970412aa446b21080daa9e7e6b017ab766c2d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-14 03:04:08.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/8/upstream-check/cinder-tempest-plugin-ontap-fcp/ddfd61e/ : RETRY_LIMIT in 8m 55s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/98d275e/ : SUCCESS in 44m 01s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/8/upstream-check/cinder-tempest-plugin-ontap-nfs/9805208/ : SUCCESS in 46m 51s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/8/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4d2db88/ : FAILURE in 44m 28s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"08cb6e9a9685f9662ac5777e80e0c1c356cf5e68","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-02-14 03:08:48.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_33609\u003e.","accounts_in_message":[{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"}],"_revision_number":8},{"id":"f5997863c8c3439c4a188aaea2d5539a3e38e8dd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-14 03:54:52.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/847131/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/55f7813/ : SUCCESS in 1h 38m 46s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/31/847131/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/9f8a34b/ : SUCCESS in 1h 58m 44s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/31/847131/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/1e6bd9d/ : FAILURE in 2h 39m 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/31/847131/8/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/55f3bd4/ : SUCCESS in 1h 45m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9e3bf33541d5d84ae66ba10cff7a01990510f1db","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-02-14 05:17:52.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/31/847131/8/check/DellEMC_PowerFlex/222374f/DellEMC_PowerFlex/1603 : SUCCESS in 1h 27m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"a85b2c9cf968774a3e5fdbb41be6c4a57c084b64","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-14 05:25:28.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/f93cfc9cc89f4d52b750011535eef9dd : RETRY_LIMIT in 1m 29s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/8a00409308784b4d84eb2e58b2bb0c94 : FAILURE in 1h 51m 01s","accounts_in_message":[],"_revision_number":8},{"id":"e4805531ea1472c5453e18560c6d22d6f2acac24","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-14 11:59:52.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/760ff2edc3dc406b9baaacd55cf3fd7a : SUCCESS in 1h 12m 37s","accounts_in_message":[],"_revision_number":8},{"id":"03e7f3cb188c664dfe62e407f8a58ac13d45b624","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-14 21:09:15.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/8/check/DellEMC_Unity_FC/2ff45da/DellEMC_Unity_FC/2271 : SUCCESS in 1h 15m 35s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/8/check/DellEMC_Unity_iSCSI/9b39dc4/DellEMC_Unity_iSCSI/2131 : SUCCESS in 1h 43m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"d25332d3cb4a92ead6798c8adb07a906037d08d4","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-02-15 08:17:01.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/847131-8/HPMSA-iSCSI/ : SUCCESS in 1:37:44\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/847131-8/HPMSA-FC/ : SUCCESS in 1:19:30","accounts_in_message":[],"_revision_number":8},{"id":"0a01e86ca3f56cc52c0d4ea8d6dfe7508135ef82","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-02-15 14:56:43.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/847131-8/Seagate-iSCSI/ : SUCCESS in 1:54:04\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/847131-8/Seagate-FC/ : SUCCESS in 1:35:34","accounts_in_message":[],"_revision_number":8},{"id":"a6f2f58a32c9ec83282e6d2f04cbfc6dc2e3a0b1","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-02-15 15:05:06.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/847131-8/PVME-iSCSI/ : SUCCESS in 1:54:08\n- PVME-FC https://os-logs.seagate.com/dell-ci/847131-8/PVME-FC/ : SUCCESS in 1:35:32","accounts_in_message":[],"_revision_number":8},{"id":"48f2b14e354d0569e862090cea147c30aa6e81e5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-18 22:14:40.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/8/20230219022947/NEC-V-ISCSI : SUCCESS in 2h 16m 02s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/8/20230219022947/NEC-V-FC : SUCCESS in 4h 44m 50s","accounts_in_message":[],"_revision_number":8},{"id":"b504d35513b4b4a6bb9ff2f2e7ed31da4537a284","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-02-23 09:47:01.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/8/20230223115134/NEC-ISCSI : SUCCESS in 3h 32m 58s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/8/20230223115134/NEC-FC : SUCCESS in 6h 55m 24s","accounts_in_message":[],"_revision_number":8},{"id":"8d56dd8d99ce624f61933574e1ffae19f96a7c5d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-01 08:09:12.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"44747ca99e474e0279312ee99acace2589807d3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-03-01 12:49:08.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"6d72788fe406f0d215e293690c35267f90290472","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 13:02:27.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01725bb85d014e45bc48527210e19ae8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d92f228c311d4ff1951bd82238cf61c8 : SUCCESS in 9m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3ffdd511ce5b418cbcced37e92609d67 : SUCCESS in 11m 36s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"813771adfdd00ad51ec814659b37d43a1dd670f1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-01 13:17:45.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/847131/9/check/iscsi-driver-alletra-6k/bf6290c/bf6290c.html : FAILURE in 26m 06s","accounts_in_message":[],"_revision_number":9},{"id":"807e98620358fbc8eaac64fac6e3641c5fd1551e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-01 13:18:13.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847131 : FAILURE in 26m 07s","accounts_in_message":[],"_revision_number":9},{"id":"b0f8122088c29ffcf05cc8333189d791cc098e4e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-01 14:16:01.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9c20151744c34aa5831e94f253995886 : SUCCESS in 1h 23m 58s","accounts_in_message":[],"_revision_number":9},{"id":"a7ec4bf6ae5e6a19598aa90304f302fb0c180b55","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-01 15:11:12.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/847131 : FAILURE in 23m 42s","accounts_in_message":[],"_revision_number":9},{"id":"20ff07b08bb2e0cd8bc20fe42d4b1035b0ee27cc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-01 15:51:21.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/3dcb6d1/ : SUCCESS in 1h 53m 02s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0ac2978/ : FAILURE in 2h 20m 00s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/60d632e/ : SUCCESS in 1h 38m 05s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/8fd5d87/ : FAILURE in 2h 22m 02s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/835d73f/ : SUCCESS in 1h 55m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ca2c11ad8827684d33e31f52f926ea85544f1544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 16:04:42.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffcb77ba0d844b5db9807d4525d9d538\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3271dd387ed41f2953cd366b2e95151 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e705ce33e9344f2198911c74b6f7f775 : SUCCESS in 11m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef606b2460e14618be8fc69110f6330f : SUCCESS in 12m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4438d30c8bd41e1a54ccd3852692cb9 : SUCCESS in 13m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/70d6a4d63e4b46738006b96328650e35 : SUCCESS in 1h 06m 08s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/12e31a547364443e9880749b8b492311 : SUCCESS in 56m 13s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0ec3215b83544e7eab6460f26e331e48 : SUCCESS in 1h 31m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd425b608e074516a1365d714bc64e09 : SUCCESS in 57m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/de4a746e6d0a4a759ab39c9fd517e8ec : SUCCESS in 17m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/303f1498c455421e81a396461c1630ea : SUCCESS in 6m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/849df6310b464c068fab4e7179385170 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/67fe3811874a426083f9d51bd6e222a6 : SUCCESS in 8m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9e0071e129e24a93839c3b25398dff54 : SUCCESS in 6m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ac60f9a1f61c44c39edc8ce44dbceb9d : SUCCESS in 43m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d9d0f8acc7c147469ade3ff41ac80142 : FAILURE in 8m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/291020ba2fa74289a94b8c54fc2f245f : SUCCESS in 1h 16m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/591a02713d534662a1b2d2d52217c1a9 : TIMED_OUT in 3h 07m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d30f24ad052c48ee9eaae7bb53746e76 : FAILURE in 1h 45m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/55779a4515e54eab9c39f8b16f566d76 : SUCCESS in 1h 12m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/09ef95e1d2854234a227a532a3524046 : FAILURE in 1h 32m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dbfde11d746a47408ccbde8133659292 : SUCCESS in 36m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3d0c25914b3d4070ac827bbf6322b739 : SUCCESS in 1h 10m 27s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3ddd7bd18ea14013ab9ae39284c81104 : SUCCESS in 2h 10m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23a099693ed94a45a4ad07f3f25b1157 : SUCCESS in 42m 29s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d8492cb01e6418d8fa2ea206fa36a34 : SUCCESS in 1h 12m 25s","accounts_in_message":[],"_revision_number":9},{"id":"f165dd325b3ca061fe5b90d8de6de94bb07441d2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-01 17:47:18.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-847131-9/ns5_nfs : FAILURE in 4h 20m 37s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-847131-9/ns5_iscsi : FAILURE in 4h 51m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"74becec24f2b871146781afa4b2123a38c86d196","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-01 18:49:48.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/c3627caf25734f10b38f4374626c2f25 : FAILURE in 1h 33m 45s","accounts_in_message":[],"_revision_number":9},{"id":"2a018ca4234c3b665c4e21c864e9f484e4788ff3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-01 23:06:38.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/9/check/DellEMC_Unity_FC/013ed7a/DellEMC_Unity_FC/2392 : SUCCESS in 1h 31m 17s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/9/check/DellEMC_Unity_iSCSI/e396cf2/DellEMC_Unity_iSCSI/2244 : SUCCESS in 2h 07m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"edc304b361da0c64d4a0ade4e475d1f7db753dd3","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-03-01 23:15:46.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6ce6123d5f1020bb6f940e2aa7545c662743e28f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 23:38:23.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c0cb49450f344b0a624747c368fb66e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7815dfdd75674ae2bf3bdd560253438b : SUCCESS in 20m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/491ea8fb2e3544e4a7dd501205c80cd2 : SUCCESS in 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3d44c3afbb5cc8cff9a2b57b58596a8f83827a0f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-02 00:30:43.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ce9ee3ae13134e57a8cdff14940a8d1a : SUCCESS in 1h 13m 20s","accounts_in_message":[],"_revision_number":9},{"id":"1ba2f3a4fde5798c11d036213cfccb116c34ce69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 02:01:18.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc39327f77914bb18f6dd6d980dd9de0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9801aa0c200a471db5e77c5214585d18 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5aead0f46aa495cb4c17724c8bf12e9 : SUCCESS in 9m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ddf81248d3344bc8e62a018947dbc3a : SUCCESS in 14m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfbbd833237749b1b8db52dbe9444998 : SUCCESS in 13m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/7507ba1201364979a37c6697c9c33503 : SUCCESS in 1h 05m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/eb2e7112f70941c5875f859169e39ddb : SUCCESS in 53m 06s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/784bb4d189e94c88ae453a5a021e6dc1 : SUCCESS in 1h 18m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/827010ea1af54ef4a16a3e16cdedd0b2 : SUCCESS in 40m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c0e6266106884008b086a537712225bc : SUCCESS in 15m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cf69a1ceaaa44a85bd46fce3a76e1a8f : SUCCESS in 4m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d7d9f8d419734781bfd1554b6dd9bb21 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ea75f8789402450c9b9b77dade7836e7 : SUCCESS in 5m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cb989bd0b94d4ac5b0a458c84014e882 : SUCCESS in 5m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/25c0b2f5f78549818f0180024306417b : SUCCESS in 52m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0dbacf64a8924ae395c985f3df512267 : FAILURE in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c8b0d484a4704faaa308cdd3ef04a829 : SUCCESS in 1h 20m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/69a9cda10f1f48eeb529789a53e649b2 : FAILURE in 1h 18m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1ff4e6e3fc5441c7be714c586738c30a : FAILURE in 1h 46m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3f6e79382e074c8bbaa2ae10056f40dd : SUCCESS in 1h 10m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9e03529a5e1d4862bf37441157260e33 : SUCCESS in 1h 18m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/18860c155ea14be0916df4a79db1b08a : SUCCESS in 51m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c91c564e8cd148dcaa3e6c5776596a27 : SUCCESS in 1h 09m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ab80b64219924c1db7ab37500f5464df : SUCCESS in 2h 38m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d5a7d7befe34cdeabd1234e5682dcc8 : SUCCESS in 58m 19s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f9d96666b5c4a158458555af6a38621 : SUCCESS in 1h 20m 30s","accounts_in_message":[],"_revision_number":9},{"id":"3ed59f030abf36c0f964203be35c488de137ce5d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-03-02 03:39:12.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/847131/9/ : FAILURE in 55m 42s","accounts_in_message":[],"_revision_number":9},{"id":"87d93b703590ce70ce500cb4061e3945b6f5aa7a","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-03-02 06:50:52.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9cfcfe4f35e366fd83f9101e4f99ba2ee1534607","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 07:00:07.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64f4702481a9441f8ec0e52e1e32d7b6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eae6d92c60e94b1fbd0068071df99c3a : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/13c6cd7f35294a9ab060c785f05122e2 : SUCCESS in 8m 05s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6a76edffcba67d17ff4e4303f91aaa8603401712","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-02 08:02:04.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/98b8adb17a4e497b92832462b0d0e972 : SUCCESS in 1h 09m 33s","accounts_in_message":[],"_revision_number":9},{"id":"4d94eb18274e4033f124ef0198df7f7623348bfc","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-02 08:25:29.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/31/847131/9/check/fujitsu-eternus-dx-iscsi/23c2b38 : FAILURE in 46m 21s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/31/847131/9/check/fujitsu-eternus-dx-fc/48d0ac4 : SUCCESS in 1h 33m 38s","accounts_in_message":[],"_revision_number":9},{"id":"c3e8615050ff18357cd5d497881738d6deb275dc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-02 08:50:26.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/86a289a9d5ae4e97938829e5eda5eb3a : RETRY_LIMIT in 1m 03s","accounts_in_message":[],"_revision_number":9},{"id":"64a9e7365739ad3a2c31f799d013abc2bd5bac35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 09:17:08.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9147524cccde4d0cb709ea352abd6fe4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b3ecfb8cff144c1925940c347f5fc15 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b61aef2f77af428189b8180981c67c34 : SUCCESS in 10m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ec6c57c95a948fabd111f91d3c8c4c9 : SUCCESS in 12m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2b96598de2d4859825d37281cdcac19 : SUCCESS in 13m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/056b9e5476ea45f3ac73263c15a449a4 : SUCCESS in 55m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3e3fdaac49dd4aa4b0ccee03446148b2 : SUCCESS in 1h 03m 38s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/abce6c902d9c493ea415dd08eff8a08a : FAILURE in 54m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbd56699a26c43fe950264e4b41469ea : SUCCESS in 47m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/74f9f47c2e394d4f8776015b075f2ff3 : SUCCESS in 15m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b82979e6edde4f14af6cdcdae6aa921a : SUCCESS in 4m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/85a258ad576040daa90e467f6c1c9da5 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/51e0bd68a9e44dcdb9325c96ebccc5eb : SUCCESS in 5m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e35e9a0282fd4ebfba9f6c3b29637480 : SUCCESS in 5m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b2a7f5be98d44648b980bfe9fc43ccc4 : SUCCESS in 31m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cfdbf53535a14de79d377e043bd2ad73 : FAILURE in 6m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c31934a0a1854004a83253afbc3430b0 : TIMED_OUT in 2h 04m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2b3dd26d626e46a1873b38b14d4d3b33 : FAILURE in 1h 39m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e1a522c16ac4473fa3e3c52429e1e20f : SUCCESS in 1h 49m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e9a4375383464f3b8340a3524cdb1480 : SUCCESS in 1h 08m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/da1bc417e04e44ca95edc7dad7774382 : SUCCESS in 1h 33m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f8db06f12fa247e385a2dd1f05b19195 : SUCCESS in 29m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ce2d9fa95ee445a48be9f3aae91536ba : SUCCESS in 1h 02m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2138be13fa03410191060d4623738a9f : SUCCESS in 2h 24m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88aa04a2bff84afaa8759f845ead50ba : SUCCESS in 1h 07m 24s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1191088b31f04194b41bb00279e5a972 : SUCCESS in 1h 24m 42s","accounts_in_message":[],"_revision_number":9},{"id":"a1fdb2a30bb6960ff6ee3867fb85f208438cf956","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-02 12:17:38.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/87c61e4/ : SUCCESS in 1h 47m 45s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3c8325f/ : FAILURE in 2h 03m 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/4d4f0b5/ : SUCCESS in 1h 51m 45s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/da23c47/ : SUCCESS in 1h 51m 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/08efddd/ : SUCCESS in 1h 43m 42s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5f5e231357c8f15c93492e3df609f3ce11f24441","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-03-02 18:55:18.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"467cced0b044a00e7b3e741e4cdc6f280f704db0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 19:36:03.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aae929966fc641a0b97cb7b407fe1b1c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/83d57c20b2ec4722a7c28803240e2178 : SUCCESS in 37m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7f8e2d57838c4522af99b6d75465a7dc : SUCCESS in 34m 45s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ed66511cd01750614ed7c59a83622cc77912d910","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-02 20:08:52.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/360b05e3ba35459c932be62d94d0d0fb : SUCCESS in 1h 11m 05s","accounts_in_message":[],"_revision_number":9},{"id":"fe6f58599dbd484471ea1136833913b42eaeb6e7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-02 20:20:55.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/31/847131/9/check/fujitsu-eternus-dx-iscsi/3702d33 : SUCCESS in 1h 25m 03s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/31/847131/9/check/fujitsu-eternus-dx-fc/d876745 : FAILURE in 3m 11s","accounts_in_message":[],"_revision_number":9},{"id":"aac8c797413e0802ac67853e64aa064170ed8847","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-02 21:12:12.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8e3986e/ : SUCCESS in 1h 32m 31s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/50f8d77/ : FAILURE in 2h 14m 01s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/c08194f/ : SUCCESS in 1h 52m 11s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/b312f16/ : SUCCESS in 1h 58m 22s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/64ca4dc/ : SUCCESS in 1h 43m 55s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"fa64fb6dd35e72dc195fde784fff470ff929bd0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 22:38:30.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d2ff81308c3486f8b6e978795b5bf15\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/188a212f2a62471f962c1631346d222e : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0e9a165ce13425384ddb7b36ebabe4f : SUCCESS in 10m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a3f959cbc0f4d8cab8ca33a459df26b : SUCCESS in 10m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59438595511b4ed9852d140838306158 : SUCCESS in 12m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/bc429a0da7db4746952b27c171992980 : SUCCESS in 1h 04m 34s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/957609e843e54c1a9bbdebbf0990393b : SUCCESS in 1h 09m 30s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/32f4c266d5fd460ea5113ede0115fb2a : FAILURE in 1h 08m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52380ef013934011b3ff618fd23e45a4 : SUCCESS in 58m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/af0dd8bb58bc4e769a9825f98ad6db86 : SUCCESS in 15m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/acc797bea4bb48a78b5e8329653c05eb : SUCCESS in 5m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7961708a32024c6388ff6b33e229f646 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fcaa49d9a35646a9b430527f49e9a87c : SUCCESS in 7m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/da0d1a58de1041fd81002f865b79cbc5 : SUCCESS in 5m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d544047033ad42019ac29056ad4e7ecc : SUCCESS in 43m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a491ffd3754c409faf5b6d2709c46f05 : FAILURE in 8m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/88bb426dac7046c9a5011dc06603aa90 : SUCCESS in 1h 22m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/457566c7e14246748dc60bcf950c5f7a : FAILURE in 1h 34m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/21165c6055584fe1832bd22c0a47515f : SUCCESS in 1h 49m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/995158433be04a0a9e5a7c8955b9a682 : SUCCESS in 53m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/487aba1b702c49858d4c1b1cb9d6075d : FAILURE in 1h 28m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9e2546bcb4034852a1e0adb1f352d16e : SUCCESS in 35m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e6686678d3ab4f329d5e02ebedc8ee87 : SUCCESS in 1h 15m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a0f9e0ce54f84d4f840035384971139e : SUCCESS in 2h 05m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7866d048be1a498fb86082b356b7264b : FAILURE in 50m 10s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a87e18b4a6f7421db44d1247db8ccba2 : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":9},{"id":"bf514be1073f3eb2ca0f1b058c1f04b603bfbea6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-03 03:06:05.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/ad71957884fd4b229899f6b33caa39d3 : SUCCESS in 2h 10m 37s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"7d27294fecacbdb71deb479f2b4f43ba213b16d3","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-03-03 23:44:32.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ee3204b7894d993ce5da46f15aece73457ec45a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 00:05:44.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac6212ec09084319bed6df24f6bd670c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f54aa97140347c7be5c969740edf97c : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0e0521704e5c4923b5a69b3d1e7f14b6 : SUCCESS in 19m 14s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"57fac8f81f274a7bfefb6a1aa6ca6dfa0c5306ee","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-04 01:02:15.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/92179625a5b949f8a157cf071067fab8 : SUCCESS in 1h 15m 53s","accounts_in_message":[],"_revision_number":9},{"id":"65c3a7e7a0e30cb3ca8e6df46b8e93f271f5d4a0","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-04 01:06:02.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/9f587034baed466896e1972558d566e8 : SUCCESS in 1h 20m 53s","accounts_in_message":[],"_revision_number":9},{"id":"f3de8930cf5dc6ea6aa47cb5d4784f6cce880faa","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-04 01:16:40.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/31/847131/9/check/fujitsu-eternus-dx-iscsi/89f3786 : SUCCESS in 1h 31m 29s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/31/847131/9/check/fujitsu-eternus-dx-fc/542c858 : FAILURE in 3m 14s","accounts_in_message":[],"_revision_number":9},{"id":"5277249f2ef01c81b21ddf5406077838fdf113bd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-04 01:53:06.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/55c023b/ : SUCCESS in 1h 44m 10s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b7a6c97/ : FAILURE in 2h 05m 42s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/0b85dd3/ : SUCCESS in 1h 41m 31s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/af8c067/ : SUCCESS in 1h 48m 19s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/31/847131/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/961ca5d/ : SUCCESS in 1h 44m 52s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2328561aaf8deb31fe91ab9bcc1c71d79e834d65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 02:23:01.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c41dd1a7b054738af300529b9ea63a8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75b19bc8d0fe40f889c43f9929184273 : SUCCESS in 13m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fcf45b39aa749999380b76c54e35afd : SUCCESS in 12m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5518e964d79941a5b6c0c18950673b49 : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5317894253fa44bdbe5764e86f46e700 : SUCCESS in 12m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/5cf82ede0af1449b82eeed2c112f5756 : SUCCESS in 1h 13m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4c61649dcc2a4e5aaa9f8c15de225522 : SUCCESS in 59m 37s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/afebd9061b0e4ecfa30fb7069add84b7 : SUCCESS in 1h 53m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18a9ee3b31df422490000c6df17f6779 : SUCCESS in 1h 01m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d3aac33c3a394b0ab260b95cb0287b23 : SUCCESS in 19m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5272b82bf5574f428fb184233e3ce9f5 : SUCCESS in 5m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c2b1fd002acb4b7b9539986ec8851333 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/035cb13b121c4b27966a309bf10e075a : SUCCESS in 5m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f291403308214837b5dc74870ecf4af3 : SUCCESS in 5m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a51a582a85de461cae7aa40aed28f331 : SUCCESS in 56m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d0cebeb6c0c94f22af5a0ce5f86858da : FAILURE in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bb54015011104b1dbc13bfdc81e1962c : SUCCESS in 1h 24m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/75d8d50804044a979bc15093b1d42b87 : FAILURE in 1h 35m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/520ec19d0f9040d39d7993b4ff6ce32a : SUCCESS in 2h 07m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/24004cb0e0ae4c3d9c47a7b6267e51ce : SUCCESS in 1h 19m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/256e4960e1eb404ea70b995d02caf99e : SUCCESS in 1h 47m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/76451731be4e4191b4a713a1f75c908c : SUCCESS in 40m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9099658062f74ff0b6b08e8ab6696cb4 : SUCCESS in 1h 25m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/edf037b207964d41b828e207d0f244f3 : SUCCESS in 2h 28m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a7762c49999482cb1fa3c2ba8caf678 : SUCCESS in 51m 38s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17322946fd8142c9aa84b239efee1238 : SUCCESS in 1h 22m 41s","accounts_in_message":[],"_revision_number":9},{"id":"bed1387ca4637c356d35bc02cd5f7faa84f00812","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-03-04 02:30:48.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/04_03_2023_02_23_09\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"68e6abb89b2dd8968e088f20f1a40b609692bea9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-03-04 03:23:55.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/9/upstream-check/cinder-tempest-plugin-ontap-fcp/3081fb6/ : SUCCESS in 1h 00m 11s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/9/upstream-check/cinder-tempest-plugin-ontap-iscsi/b452ea0/ : SUCCESS in 44m 57s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/9/upstream-check/cinder-tempest-plugin-ontap-nfs/2baddff/ : SUCCESS in 43m 31s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/9/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/f663da3/ : SUCCESS in 46m 01s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"b8a7046f94882ef0749c4075897ea6828b877119","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-03-04 03:56:38.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/9/check/DellEMC_PowerStore_iSCSI/e14307b/DellEMC_PowerStore_iSCSI/1094 : SUCCESS in 1h 14m 42s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/31/847131/9/check/DellEMC_PowerStore_FC/4153e0f/DellEMC_PowerStore_FC/1216 : SUCCESS in 1h 29m 10s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/31/847131/9/check/DellEMC_PowerStore_NFS/3685dc5/DellEMC_PowerStore_NFS/733 : SUCCESS in 1h 16m 45s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/31/847131/9/check/DellEMC_PowerStore_NVMeTCP/8d0bdee/DellEMC_PowerStore_NVMeTCP/597 : SUCCESS in 1h 27m 12s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"86721ab8475bfab07aa576df20990743950189ea","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-03-04 04:03:53.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/31/847131/9/check/DellEMC_PowerFlex/2b99582/DellEMC_PowerFlex/1699 : SUCCESS in 1h 36m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"f0b5e554dec45e6de8e6d7483432c4f7b3a889bc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-04 04:07:43.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/d12380c055b647948d8ec4dd9a49f586 : RETRY_LIMIT in 1m 06s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/8c806366609f47b6b618a9f5a7d763a1 : RETRY_LIMIT in 1m 53s","accounts_in_message":[],"_revision_number":9},{"id":"2f1947fc99c54fb13efc8b480e6c52cdee60477c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2023-03-04 04:21:52.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/31/847131/9/check/DellEMC_SC_FC/a09e555/DellEMC_SC_FC/1165 : SUCCESS in 1h 12m 42s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/9/check/DellEMC_SC_iSCSI/9364f27/DellEMC_SC_iSCSI/1148 : SUCCESS in 1h 54m 24s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":9},{"id":"91cc7efcb2bcf00a2034de8ad80a13e1c68eeee4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-04 16:18:42.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/9/check/DellEMC_Unity_FC/3569e87/DellEMC_Unity_FC/2429 : SUCCESS in 1h 37m 17s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/9/check/DellEMC_Unity_iSCSI/195d935/DellEMC_Unity_iSCSI/2278 : SUCCESS in 1h 42m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"72eb12f71ef7f6a7dbc405f301d2522899727577","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-03-05 16:21:12.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/9/20230305222616/NEC-V-ISCSI : SUCCESS in 2h 38m 49s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/9/20230305222616/NEC-V-FC : SUCCESS in 2h 54m 52s","accounts_in_message":[],"_revision_number":9},{"id":"4ca524628b729087a40a39e59d87a3053203eaf7","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-03-06 23:07:47.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/847131-9/HPMSA-iSCSI/ : SUCCESS in 1:34:23\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/847131-9/HPMSA-FC/ : SUCCESS in 1:21:11","accounts_in_message":[],"_revision_number":9},{"id":"b719e7f4c3be80577b2ea8e10ba6015baa9acdc1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-03-07 01:30:08.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/9/20230307032138/NEC-FC : SUCCESS in 3h 28m 14s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/9/20230307032138/NEC-ISCSI : SUCCESS in 7h 08m 26s","accounts_in_message":[],"_revision_number":9},{"id":"d2cc42b452804ee8d2156c783eb43f03e704d872","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-03-07 09:43:39.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/847131-9/PVME-iSCSI/ : SUCCESS in 1:47:18\n- PVME-FC https://os-logs.seagate.com/dell-ci/847131-9/PVME-FC/ : SUCCESS in 1:36:35","accounts_in_message":[],"_revision_number":9},{"id":"bba624c9c9c66c29de62531cb19d6311db0509e9","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-03-07 14:04:17.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/847131-9/Seagate-iSCSI/ : SUCCESS in 1:44:46\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/847131-9/Seagate-FC/ : SUCCESS in 1:36:31","accounts_in_message":[],"_revision_number":9},{"id":"31a42832cbb1abd212106f782964f630eba5817e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-21 08:06:34.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/31/847131/9/check/iscsi-driver-nimble/97186e0/97186e0.html : FAILURE in 1h 58m 13s","accounts_in_message":[],"_revision_number":9},{"id":"21e9b359b9f8d4bdbe4067fbeac6d58b6697802a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-05-08 19:32:02.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"df264b95cd0e65bbec4e0da2ce0d159be0dfc491","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 19:51:20.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccdd4ac65339430da47c3f3c501ed280\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55f11e30afbc4289905f27b0071db2a5 : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2cda5558cbc44b7083bb3354bf3b620f : SUCCESS in 17m 15s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e09a5e08042ac1f572dac661f9a5f11f2dcd95cc","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-08 20:32:20.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/f9bf64a1f020408aa4dd8a34684916b5 : POST_FAILURE in 57m 00s","accounts_in_message":[],"_revision_number":10},{"id":"1480782a3673e367e74b221dcc6b57a8ffadd7d0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-08 20:51:30.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/847131/10/check/iscsi-driver-alletra-6k/a06ea69/a06ea69.html : FAILURE in 1h 17m 40s","accounts_in_message":[],"_revision_number":10},{"id":"f93097badf013e089256280ea8c03bc764faa9a2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-08 20:58:46.000000000","message":"Patch Set 10:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/78b64919e9924adf96e7e0105d5470c1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/03d681da6eb64dba8d845330094cdc9f : SUCCESS in 1h 05m 46s","accounts_in_message":[],"_revision_number":10},{"id":"7f3247bc15d899c8b579285ab680253ed47d87b1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-05-08 21:07:41.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847131 : FAILURE in 1h 32m 22s","accounts_in_message":[],"_revision_number":10},{"id":"cff22eb69bfd38c4796d6696626e32b85dc49d8e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-08 21:38:44.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/10/check/DellEMC_Unity_FC/f9a086b/DellEMC_Unity_FC/105 : SUCCESS in 1h 40m 42s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/10/check/DellEMC_Unity_iSCSI/e2c59b9/DellEMC_Unity_iSCSI/108 : SUCCESS in 2h 06m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"f21d6081110946926bf123d851ffa857568b39ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 22:03:33.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e96e7c838d4e4e96a8cc04ff583ce9ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d62d058613dc437c913bf8333991125e : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f842a2c04ba4344a3560f2f39e9d35e : SUCCESS in 12m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/851fee112806421183fc46fb0faa8d55 : SUCCESS in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d24ceca2fde40fc9e7d274a047d9eac : SUCCESS in 13m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/ac36dd2a14c4476e9cea1fb2ff8356f5 : SUCCESS in 39m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d18a5e4bde4e4f2f94bff6a561105e21 : SUCCESS in 1h 28m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/422017397a3a40adbbdf07ca58ea2666 : SUCCESS in 45m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4401e8aea2a4466dba8b700928def35f : SUCCESS in 15m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b2dc4e4c21fb4a77bd73990234948e46 : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a4e2375d95b340a5b03058596af13f3b : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2f0de9f02c34405b879bd3d92d852732 : SUCCESS in 5m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9a4b06d6b4a4466eae53a57b453a164b : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/177a880c8a1e43b3b2e6ad0e7200660b : SUCCESS in 45m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/58123b9b575f4d35a4d8a0ca0cbc5ac4 : SUCCESS in 5m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f85c2ec6a02042b2899469ae1fec34ff : SUCCESS in 1h 56m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/840a21f3de7048819e81f335a11398ae : FAILURE in 1h 48m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f19ae5bd139f4c848758182dbc84ad6c : SUCCESS in 1h 54m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ee62f6b7d5c24bbf925ec09efb933c1c : SUCCESS in 2h 25m 16s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9b3c99a12a294ea0836025b2fe0152bd : SUCCESS in 1h 05m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/34ae2c02e27c44ffb0b1bb8cdc03558d : SUCCESS in 1h 26m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fb3b25723e9b4596a4d9a3f82573367e : SUCCESS in 41m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8df914c64a8b475384865a4bde0ed75b : SUCCESS in 1h 14m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/767510c9c8bf434fb9f2daeeeaa89e06 : SUCCESS in 1h 40m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3f48e76651ce496ab1ee7a443ef7cd95 : SUCCESS in 2h 25m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21b4d9ee3a6c45c88863f088e4ceb095 : SUCCESS in 38m 45s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4397efd1cf3946d8b85079653ec309aa : SUCCESS in 1h 31m 37s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/efb93c38d7f74b46936e6a4867fcf87f : SUCCESS in 10m 14s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6114f45e3dfd74860ea60be581103a4ed1e387c9","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-05-08 22:05:25.000000000","message":"Patch Set 10:\n\nBuild failed.","accounts_in_message":[],"_revision_number":10},{"id":"7348e359fae75f1173539530d6400b41ef9fa5bc","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-05-08 22:18:54.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-847131-10/FC/ : SUCCESS in 2h 22m 29s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-847131-10/iSCSI/ : SUCCESS in 2h 46m 29s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"b57b3a212b4d8c9ed98e3f9933c00f3a88ef29e0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-08 22:50:20.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/10/upstream-check/cinder-tempest-plugin-ontap-fcp/973147c/ : RETRY_LIMIT in 11m 14s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/10/upstream-check/cinder-tempest-plugin-ontap-iscsi/dbc1039/ : SUCCESS in 43m 24s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/10/upstream-check/cinder-tempest-plugin-ontap-nfs/f3e8598/ : SUCCESS in 45m 49s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/10/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/9c00922/ : FAILURE in 44m 22s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"c2aad327afef8a1c7e95ae9fbe1568d8f0b9ae31","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-05-08 23:03:01.000000000","message":"Patch Set 10:\n\nBuild failed.","accounts_in_message":[],"_revision_number":10},{"id":"f56d9864eacf95c54342149c9ba20e40d25c719a","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-05-08 23:03:32.000000000","message":"Patch Set 10:\n\nBuild failed.","accounts_in_message":[],"_revision_number":10},{"id":"e0955ac264a1eb256406c692ee97b193d7d93d23","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-08 23:24:36.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-847131-10/ns5_nfs : SUCCESS in 3h 42m 52s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-847131-10/ns5_iscsi : FAILURE in 3h 52m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"03c12ec1707decc5325c7dafc59738bc95685183","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-05-09 02:16:12.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/10/check/DellEMC_PowerStore_iSCSI/362d246/DellEMC_PowerStore_iSCSI/26 : SUCCESS in 52m 26s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/31/847131/10/check/DellEMC_PowerStore_FC/acff913/DellEMC_PowerStore_FC/27 : SUCCESS in 1h 07m 26s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/31/847131/10/check/DellEMC_PowerStore_NFS/0f50e11/DellEMC_PowerStore_NFS/25 : SUCCESS in 57m 03s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/31/847131/10/check/DellEMC_PowerStore_NVMeTCP/9d862b7/DellEMC_PowerStore_NVMeTCP/25 : SUCCESS in 1h 01m 46s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"a69f8def62bedd1785437fb81870b29f72f0afb0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-05-09 03:02:15.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/31/847131/10/check/DellEMC_PowerFlex/33c388e/DellEMC_PowerFlex/69 : FAILURE in 1h 43m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"7cc436a5d454f1333697d6a7a9d95d3a1727ad67","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-05-09 14:59:41.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/10/20230509211325/NEC-V-ISCSI : SUCCESS in 2h 27m 08s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/10/20230509211325/NEC-V-FC : SUCCESS in 2h 46m 13s","accounts_in_message":[],"_revision_number":10},{"id":"b7e49620163ff7fc87524ac37020acd3ca6ee700","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-09 17:51:36.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/10/check/DellEMC_Unity_FC/15ee624/DellEMC_Unity_FC/115 : SUCCESS in 1h 35m 16s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/10/check/DellEMC_Unity_iSCSI/2bef976/DellEMC_Unity_iSCSI/118 : SUCCESS in 1h 54m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"8d69279f05753974b3c5a920d704157bb564850e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-05-09 23:29:30.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/10/20230510051012/NEC-FC : SUCCESS in 3h 18m 25s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/10/20230510051012/NEC-ISCSI : SUCCESS in 3h 19m 15s","accounts_in_message":[],"_revision_number":10},{"id":"eb2bc7da6f1c4f0c51e126a7bd911d7edaa55599","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-05-10 02:52:52.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/8b0e6860e240445c89c75b0c2adb8c52 : RETRY_LIMIT in 1m 00s\n- devstack-plugin-ceph-dsvm-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/31/847131/10/check/devstack-plugin-ceph-dsvm-focal-py3/34f77fd/ : FAILURE in 58m 36s","accounts_in_message":[],"_revision_number":10},{"id":"48c9ffd2672a253e8a9d9962a8f672fa619413f7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-10 09:27:13.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"26538a900fce8d3b5a100400401839d71e36ec80","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-05-13 11:21:26.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/847131/10/ : RETRY_LIMIT in 11m 05s","accounts_in_message":[],"_revision_number":10},{"id":"715c0757e4033ac559f42c407186a863d27b0f0a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-15 15:28:28.000000000","message":"Patch Set 10:\n\nBuild failed (experimental pipeline).\n\n- cinder-storpool-tempest-iscsi https://spfactory.storpool.com/zuul/t/local/build/b517a65678294b878d6df9bb1daeb4a8 : RETRY_LIMIT in 8m 11s","accounts_in_message":[],"_revision_number":10},{"id":"864a5ee6e90075634272ba4de996bb7242d2e45d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-15 15:41:21.000000000","message":"Patch Set 10:\n\nBuild failed (experimental pipeline).\n\n- cinder-storpool-tempest-iscsi https://spfactory.storpool.com/zuul/t/local/build/3d49236648344abe89dc8b2e351a086a : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"83b5fd70f60d9d30e2b4b0c2685a6fde364d1ac5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-15 16:29:19.000000000","message":"Patch Set 10:\n\nBuild failed (experimental pipeline).\n\n- cinder-storpool-tempest-iscsi https://spfactory.storpool.com/zuul/t/local/build/f5e905be606f4c20aa2f329c6f03b10b : RETRY_LIMIT in 13m 51s","accounts_in_message":[],"_revision_number":10},{"id":"5577c7323f9c653e5e334458f7224df29f9f59ab","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-16 14:05:43.000000000","message":"Patch Set 10:\n\nBuild failed (experimental pipeline).\n\n- cinder-storpool-tempest-iscsi https://spfactory.storpool.com/zuul/t/local/build/4e559315a9914cda89ef6bea2e898018 : RETRY_LIMIT in 13m 32s","accounts_in_message":[],"_revision_number":10},{"id":"010c0f385ee6a05d6ca4b738ba76bc214b2854de","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-17 09:22:17.000000000","message":"Patch Set 10:\n\nBuild succeeded (experimental pipeline).\n\n- cinder-storpool-tempest-iscsi https://spfactory.storpool.com/zuul/t/local/build/ba0693ca64f548bb87e6df84cb9dc9e7 : SUCCESS in 1h 12m 31s","accounts_in_message":[],"_revision_number":10},{"id":"6159ca2e4b3150fa0da811d99f43b7114133ea33","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-17 15:32:54.000000000","message":"Patch Set 10:\n\nBuild failed (experimental pipeline).\n\n- cinder-storpool-tempest-iscsi https://spfactory.storpool.com/zuul/t/local/build/0e46a170d94948bf92e04819b21f5ee7 : RETRY_LIMIT in 13m 23s","accounts_in_message":[],"_revision_number":10},{"id":"77fb0ec93de3c7f5d101e780e6c52133af0f33cc","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-07-21 13:23:35.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e2592b634b9e3e1281f7a36a4a32c1bd91487d4f","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-08-01 18:24:53.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bcbb5e782ba65f099ceb94eeb6205e48f9b81637","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-08-01 18:50:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e4ae3da2fcce40049a13256297b40dc7444c541d","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-08-01 21:00:07.000000000","message":"Patch Set 10:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/847131/10/fc : SUCCESS in 0h 34m 40s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/847131/10/iscsi : SUCCESS in 0h 58m 12s","accounts_in_message":[],"_revision_number":10},{"id":"3e63cc72db67e13adbb97a86c213fb49209b4017","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-12-06 06:26:30.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"e7add839e0e8f19dec087fb3fe0fec5c503411cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 06:55:48.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70e2763cff16475e8bf71064be11b3d0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3acb1bd2693548a7a2ef94e5d6241cee : SUCCESS in 20m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/34b5265663384a18ab532bc0397f1e90 : SUCCESS in 26m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/87d038b06f654c51950e714201c137f4 : SUCCESS in 10m 16s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3ec9c5fcf1c6f0a31d7fd043b4990dd64e9a4644","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 08:48:10.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d65c69643e74588a57bb09a32d87935\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0be9d60a49a142eba7753033acf56292 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fca90ac7ccfc4f299abb89f0c8fd4020 : SUCCESS in 10m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c2e7d53d2fbf4c53a4f4775f8b04c8f7 : SUCCESS in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f49a77e38aaa4d39b5a57d40178fa684 : FAILURE in 7m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/7e332fa07d954e66a5b1c270bc91572b : SUCCESS in 58m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bca5bb1d16944c589d6f75006c119e38 : SUCCESS in 55m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a5978cdbbea64e4682554063d602d136 : FAILURE in 1h 21m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd9d39d5eca04a52ad48f037cfa67d75 : SUCCESS in 52m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7425aa2b99e246d0bb04d6fee17eea0d : SUCCESS in 19m 08s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/77f6f357bed04447a2c6ae55346439a3 : SUCCESS in 13m 29s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f97f4221ce4444d5b4b768f46fbec3ab : SUCCESS in 5m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d5a4e8047e2646df9f8f97085306afa2 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/64418ed540cf40eaacf6b02dae55abf1 : SUCCESS in 5m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/39d6499609814c34889daa23a30f50b0 : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ac627f29649049198a93d6e3f018be98 : SUCCESS in 53m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/302020c205214c4da73f2a7421bd6480 : SUCCESS in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f4ea0f029af3492d99682e436fa2e676 : SUCCESS in 1h 53m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/38694921f65143668ad68880f2e8639e : FAILURE in 2h 11m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e4278e362bc140b8878d7fb692cd2eb7 : SUCCESS in 1h 32m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2689282133c64944b6b837eba9710284 : SUCCESS in 1h 55m 29s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/68bf69a876334ba6904f2ccfed72de3d : SUCCESS in 1h 07m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/de5e3f19c57d472190f354d77d9aab96 : FAILURE in 1h 25m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7ca2cd4f247d4db29bff1ca36d5800f3 : SUCCESS in 43m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d99c4f5e51cd450d8a3e24bf705f1cc1 : SUCCESS in 1h 38m 18s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8346ef483ee4445db5a37c0c276ecd19 : SUCCESS in 1h 04m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9c4bb69a5c1546b4a5ce7710780d2f02 : SUCCESS in 1h 34m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/562f0a59b3e6466c9c0a2a71775e9f97 : SUCCESS in 56m 02s","accounts_in_message":[],"_revision_number":11},{"id":"e6a92a6c657eaa9cb3a9a9f557210df811f5d177","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-12-06 08:56:49.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/75134a4faeec4a4284e4c4a488cd0614 : RETRY_LIMIT in 1m 02s\n- pure-devstack-cinder-tempest-fc-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/5e97b8b86acd4a44a1f4b5f1c8622263 : RETRY_LIMIT in 59s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/b639d97949724a6897e7d4453a340a45 : RETRY_LIMIT in 1m 01s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/d62eb061024b4c1393cd6e6aa7f56af7 : RETRY_LIMIT in 1m 01s","accounts_in_message":[],"_revision_number":11},{"id":"75417c79d719654a11bd2aaa19758cbad4381d3c","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-12-06 10:41:26.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/847131/11/ : FAILURE in 1h 09m 37s","accounts_in_message":[],"_revision_number":11},{"id":"9c447e581de1373eaa5d5e7392474693fba4e731","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-12-06 12:24:31.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/e59be1351cf2409aab53245ff34bcd4d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f2743f3697264d198dee482277ddda6d : FAILURE in 1h 21m 07s","accounts_in_message":[],"_revision_number":11},{"id":"28e79ecb26937e74439d56dc24148b13c14a7da2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-06 13:12:38.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8c5684760acc4ddfb5779faded32a244 : RETRY_LIMIT in 5m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1f8e601b4fcc489c95c86d826bc1aa49 : RETRY_LIMIT in 6m 20s","accounts_in_message":[],"_revision_number":11},{"id":"eddc780fe3497e8ecf1a10b2f1253777e8ee29aa","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-12-06 14:32:01.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"509f94df26a2f1f4b2744309a5eff448a801cc61","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-06 16:09:05.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/31/847131/11/check/DellEMC_PowerFlex/3c0a236/DellEMC_PowerFlex/999 : SUCCESS in 1h 04m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"0956c41615ba52e5f5278514f8f5427ca770cd3f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-07 00:03:43.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2f62ec9f8cb94f0bbbfde4aece11b77c : SUCCESS in 1h 11m 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8113e702600e443a89960a67b009a48a : SUCCESS in 1h 12m 11s","accounts_in_message":[],"_revision_number":11},{"id":"255d306b9feba2529a4ccadbda3fda28b217710e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-12-07 08:22:54.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/11/20231207133655/NEC-V-ISCSI : SUCCESS in 3h 31m 06s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/11/20231207133655/NEC-V-FC : SUCCESS in 3h 45m 55s","accounts_in_message":[],"_revision_number":11},{"id":"3372e53aa4429050ccbb7a46d692465009db354a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-07 11:47:21.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/11/check/DellEMC_Unity_FC/a994053/DellEMC_Unity_FC/1053 : FAILURE in 2h 55m 17s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/11/check/DellEMC_Unity_iSCSI/8f9ac4e/DellEMC_Unity_iSCSI/1067 : FAILURE in 5h 58m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"e0faf0d5dd5456690095d48e07f2d32c12e0dd9b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-12-07 20:34:54.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/11/20231208010837/NEC-ISCSI : SUCCESS in 3h 39m 37s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/11/20231208010837/NEC-FC : SUCCESS in 4h 26m 14s","accounts_in_message":[],"_revision_number":11},{"id":"cc8e7d19639cc54b7e5c48d2c692b04139c6dda2","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-12-08 01:00:07.000000000","message":"Patch Set 11:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/847131/11/fc : SUCCESS in 0h 57m 22s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/847131/11/iscsi : SUCCESS in 1h 19m 30s","accounts_in_message":[],"_revision_number":11},{"id":"2e221959f80dc6060869c9741426ffdd63cd65c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2024-01-18 12:56:47.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"30405bf27442449141f9237c47034947ac736b29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 13:39:04.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39febcbf6a374bb7bf89120db8ab22de\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7fd4457ffad48f486516ec53493b2da : FAILURE in 38m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1c9c2f04f3b44b9d8b12b1e2cbc9904e : SUCCESS in 18m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ae36176f6840411bb337c967e6804c5d : FAILURE in 38m 57s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"06c13b82253e9cb97f0999290ced1ee87f3ae541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 15:55:40.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb9987d383b24b3d8088e25bd70a0231\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e951d1a28be540538bfaa58931de3d64 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8f4dc1decfc479098239356b3ad8cf0 : SUCCESS in 11m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7fbc5844380145d88d7145eb02ad288d : SUCCESS in 10m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c08c11b37a74add8b56597e4e34d747 : FAILURE in 6m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/6be18b8de71c48728bbfe6837d9a63b7 : SUCCESS in 1h 04m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f365c9de63894c24a1aff9196865bc72 : SUCCESS in 1h 06m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bf4ef9dc47864dfa99fcc16ccee2f0a7 : SUCCESS in 1h 18m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/840ab6803e144e14bc10e30f7b3f207a : SUCCESS in 38m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e1f63f26f87b44e988ca1ea58c582d1b : SUCCESS in 18m 25s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2816d6e712624889adf5a9473bc63c0b : SUCCESS in 14m 57s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b91770fc3a244e7b9b87d025fd9888f8 : SUCCESS in 5m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/42ea7882678f4bfca2755b87ea701fa7 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/53ffa68fcc974931a26043b941144ac7 : SUCCESS in 6m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1e451034e1ee4ff4b2624821949f1976 : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3e89984751354bb48eb268e75cf9f19e : SUCCESS in 42m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/20e479f2f7ae46768677014750e3d6a3 : SUCCESS in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7ac1f23b40104cfa87e641141df2c20b : SUCCESS in 1h 37m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/35934cb4611b4301a88c58ff96549920 : SUCCESS in 2h 25m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6577916165e3406d8fa8dffd7435a761 : SUCCESS in 1h 49m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1c2a794edf934974b03bcb3d7a301d30 : SUCCESS in 2h 10m 11s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d986ad6153014ed49b308595874ab07c : SUCCESS in 1h 09m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/066e83fb200c40089d57863ee96443a2 : SUCCESS in 1h 34m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a94229c95d8c4e2c91bfd2bcb3283057 : SUCCESS in 35m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2e6488b0de1e4cc597be6500b4d75e30 : SUCCESS in 1h 35m 17s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9309599ced6940caa0ee85bc229b516a : SUCCESS in 1h 49m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a1655757f863477aac2f7387a893543f : SUCCESS in 1h 43m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd22f217f15441448f2126ed3ccf7480 : SUCCESS in 1h 17m 16s","accounts_in_message":[],"_revision_number":12},{"id":"66b57cf3fee8972aab22e95f90ac58b593ee3da3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-01-18 19:03:54.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-847131-12/FC/ : SUCCESS in 2h 36m 25s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-847131-12/iSCSI/ : SUCCESS in 3h 6m 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"e9c129b14f56c48bdfc854bf47ecb69dbab3dfc6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-01-18 19:04:44.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/e79bd543db0f4f73bda9ddd60a01b7ed\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/170975b7c61040cfbf151648e0bf669a : FAILURE in 1h 26m 57s","accounts_in_message":[],"_revision_number":12},{"id":"58729bca74d1210b0ecd98511d4b0077fbc046c2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-01-18 21:31:54.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847131 : FAILURE in 1h 05m 41s","accounts_in_message":[],"_revision_number":12},{"id":"ef584d2ccf989b09b34c74a414a99dee88fd3ec7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-01-18 21:46:49.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a59c6850575b401cae460f6445daccdc : SUCCESS in 1h 16m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/230e5afe75a5429fbe978edf0f030910 : SUCCESS in 1h 08m 28s","accounts_in_message":[],"_revision_number":12},{"id":"70a9b2894a95d9d0f49baa8370e50950604a3de7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-19 07:52:30.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/467/.","accounts_in_message":[],"_revision_number":12},{"id":"5634236375ec48404a195e462c9abc537fb372d2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-19 14:05:57.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/31/847131/12/check/DellEMC_PowerFlex_3.6/e153834/DellEMC_PowerFlex_3.6/1252 : FAILURE in 1h 22m 30s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/31/847131/12/check/DellEMC_PowerFlex_4.5/12c3b02/DellEMC_PowerFlex_4.5/223 : SUCCESS in 1h 56m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"1210fbf1264c91b02912b4b85392c21a6b22aaff","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-01-19 16:00:07.000000000","message":"Patch Set 12:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/847131/12/fc : SUCCESS in 0h 46m 27s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/847131/12/iscsi : SUCCESS in 1h 10m 22s","accounts_in_message":[],"_revision_number":12},{"id":"56720ee8148cc8b3ebdf9b35182a9675d7d5bc40","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-19 18:25:46.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/462/.","accounts_in_message":[],"_revision_number":12},{"id":"18d3af4adad97a077e96234d688da376ee18fa74","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-01-21 22:36:39.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/12/20240122040645/NEC-V-ISCSI : SUCCESS in 3h 20m 32s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/12/20240122040645/NEC-V-FC : SUCCESS in 3h 29m 51s","accounts_in_message":[],"_revision_number":12},{"id":"301bd2f2ad3d4671d2d46c2f5d8b9d237acf5245","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-01-24 00:57:33.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/12/20240124010623/NEC-ISCSI : SUCCESS in 4h 00m 53s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/12/20240124010623/NEC-FC : SUCCESS in 8h 51m 07s","accounts_in_message":[],"_revision_number":12},{"id":"34055b6019e8be39d5c500dc484917b27a2936b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-08 16:40:11.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"dde21448fe2a419c35465f7fcd139dfe32d7d72e","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-08 16:59:15.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"60e9a091fae76eea2998c536560670d769883958","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 17:03:24.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/232b1638121a41eb835f929336774955\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f541c045ba542379226eb79856d27f5 : FAILURE in 3m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6a189257d6ff49a39ca6e0c91bf4e848 : SUCCESS in 20m 01s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"2d7b0c5edf1ac8f7ba34f2a56553c9a1a6282c95","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-08 17:04:47.000000000","message":"Patch Set 13:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/572e6bd03fdd43228f7a362ddc8649d8\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/88a5c4886c1c42a4b5d08808e15f552a : POST_FAILURE in 19m 37s","accounts_in_message":[],"_revision_number":13},{"id":"8bfdab74fd15ad085acd5fb41f06a9fb19c9329d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-08 17:21:59.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/847131 : FAILURE in 40m 44s","accounts_in_message":[],"_revision_number":13},{"id":"876afc9b0f730dc3487a811ba3392205c78dbbf3","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-08 17:22:04.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/847131 : FAILURE in 40m 50s","accounts_in_message":[],"_revision_number":13},{"id":"cc70e51ee1dc64251bbca467e19adb8f3911df19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 18:57:36.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b18e2608772846268436b0c01c4a6a74\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9f13d3ffdc54fadaeec10da86157e61 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b67efbfe1c1c431c9ddf244616e8f3ee : SUCCESS in 10m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46c165bc558e410592b5a942d4793780 : SUCCESS in 9m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/72624805ce944ae180716f537539794f : SUCCESS in 14m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fd2d5dfdb1a49caa998232d622031ed : FAILURE in 5m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/ffc50bc9245049ab877d0b45ddb365c9 : SUCCESS in 58m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2f6225319f8f4532ba899466da3efc5f : SUCCESS in 1h 26m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1bff3f31989460cabf0df282e9056fd : SUCCESS in 58m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b71d95e30801417faf42c21025fc7d0c : SUCCESS in 16m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c667bfa5a8ac46bc965acac4f75530ba : SUCCESS in 4m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5a53c2c32b034a1d879a8de3ad7cd6f8 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/83085fb5eccb4c4bbfdf68df2f602053 : SUCCESS in 6m 07s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1945cbf9b9924b94be24ad69b8dd50ff : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/68b8c92711e74f70a90d6b4ff5dc7143 : FAILURE in 29m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/85715a2af42a43caa61652fd6f37cae1 : SUCCESS in 6m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e3f17101b0d44f519bdadd60b3917449 : SUCCESS in 1h 45m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a09ac778a7764b63ad70274ee588e30c : FAILURE in 2h 12m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f6f709ace0ee4f65ba95128ea73ee411 : SUCCESS in 1h 33m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/195f5f65ce24437393f6d0f113ade3e4 : SUCCESS in 1h 57m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7909c5d21ad04aa8ad006a2fa490f2fc : SUCCESS in 27m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9bf8c9c1efc0477f9d892ce56d337a1c : SUCCESS in 1h 01m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0874cfaa93f54dc783c3497f0ba17aa3 : SUCCESS in 1h 32m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b57cc10e575e4fbc8ecfd58d3a01e4a0 : SUCCESS in 41m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/855e17ee5ebb4f9fbb6096bb50a49606 : SUCCESS in 1h 29m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8b4d8e129da9432faaf07cf2b37a8205 : SUCCESS in 1h 50m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dd4e23c0932b4d0e8eec71abc9584720 : SUCCESS in 1h 33m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59e97c2241634bc4b23c5f21ad7c1670 : SUCCESS in 40m 55s","accounts_in_message":[],"_revision_number":13},{"id":"93a46d41eb7c6677d84de8e3ef39b53c4017b1b4","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-07-08 20:00:07.000000000","message":"Patch Set 13:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/847131/13/fc : SUCCESS in 0h 43m 22s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/847131/13/iscsi : SUCCESS in 0h 56m 41s","accounts_in_message":[],"_revision_number":13},{"id":"32ecd9de8ea11663c76e06752ffd380c7c2367d7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-08 20:40:52.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/db925a2362344b4ab7af768b0ff2a3ec : SUCCESS in 1h 20m 25s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8f10e34a3db14f9eaef790091afee615 : SUCCESS in 1h 12m 53s","accounts_in_message":[],"_revision_number":13},{"id":"380cb827dfffafc941eb62548c1cea5e43224c57","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-08 21:27:17.000000000","message":"Patch Set 13:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/698/.","accounts_in_message":[],"_revision_number":13},{"id":"e10ba900085fb2678a4b0269b7d8fc1b9e2f3e93","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-08 22:35:37.000000000","message":"Patch Set 13:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/202/.","accounts_in_message":[],"_revision_number":13},{"id":"c3760f173b3c39eeccaa5deffd3237d056a40c5a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-09 00:43:11.000000000","message":"Patch Set 13:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-847131-13/FC/ : FAILURE in 8h 2m 32s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-847131-13/iSCSI/ : SUCCESS in 3h 34m 11s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":13},{"id":"b6d5ff4c8f58358c784c250ea90cc004327b32df","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-09 01:42:07.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/31/847131/13/check/DellEMC_PowerFlex_3.6/9fa271e/DellEMC_PowerFlex_3.6/1677 : SUCCESS in 2h 39m 26s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"0ce5bf8ce5e0372b4173db966682137962811801","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-09 01:55:42.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"be3ed1d9b9382da65bf7320e72127654a7c94eae","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-09 02:38:23.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/13/check/DellEMC_Unity_FC/ce98d7a/DellEMC_Unity_FC/2192 : SUCCESS in 1h 20m 12s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/13/check/DellEMC_Unity_iSCSI/0526404/DellEMC_Unity_iSCSI/2201 : SUCCESS in 1h 47m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"2cb2b5992c8f41216cd5cc86f16be8f79d85b1cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-09 05:17:44.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"e782e9edd083b27c348a8033d617a7c84d3c29ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 05:26:39.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e778c5c43a2482c90a6872f62231900\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a8126503f9344b2f917c77ee2f24851d : FAILURE in 2m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/89c3781811c649199767d8b98d134934 : FAILURE in 3m 31s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c681b58e0454a63e4858fdb5bd49b69966a69b35","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-09 05:43:05.000000000","message":"Patch Set 14:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/7818068c99e44319be8d38d5d3c35e0b\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/48b1f9787b21487795e36621cbfe7a4b : POST_FAILURE in 22m 27s","accounts_in_message":[],"_revision_number":14},{"id":"068a0f0a14f411b359d82a4e55275900a8df6e25","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-09 05:44:55.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4a417fa8219a49bba96c8dbe19dbb9be : RETRY_LIMIT in 7m 06s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3a24c98fcda54a12884e780625337dd1 : RETRY_LIMIT in 7m 24s","accounts_in_message":[],"_revision_number":14},{"id":"5897bf4fdac75e1e56230156ef7ab1e5330478f0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-09 05:59:14.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/847131 : FAILURE in 40m 30s","accounts_in_message":[],"_revision_number":14},{"id":"c5e5606d9cb3758eae0758782dbc10d48ee2a8e6","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-09 06:05:19.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/847131 : FAILURE in 41m 57s","accounts_in_message":[],"_revision_number":14},{"id":"fe0f90c6636e0a5048526100b97d8f575ae64b33","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-09 06:07:24.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847131 : FAILURE in 48m 34s","accounts_in_message":[],"_revision_number":14},{"id":"6c06950b918b590f67dd51e219132a93e547108e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-09 06:52:50.000000000","message":"Patch Set 14:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/701/.","accounts_in_message":[],"_revision_number":14},{"id":"8d28422b9d77eeef219fe814042e4fa39c978df7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 07:20:20.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78cfb1232abc40f0b77b3a97eacdda04\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efc48c397c704467bfbce85d46a269ee : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/914a7f0a187f4cea9bbc8b9d6af9a196 : SUCCESS in 8m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a25864f9bc44475a97b0baa2582e7a6c : SUCCESS in 8m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/20384de1a3ca426b92d16d72919ff82c : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad4adb6138c24bccac98d9dc245ed4f7 : SUCCESS in 10m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/440280d9e72842ceb71af4744656f85a : SUCCESS in 1h 00m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/793a7a79a2004babbf43d0a2186e5b23 : SUCCESS in 1h 18m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9136b4738f8c49598a1c13ae9075beb3 : SUCCESS in 56m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/41ecb4bee8a7451d9da53f44c90c1998 : SUCCESS in 18m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7ebb8a1577494d5687f0cc3011afbcf7 : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/080ddb7ace224180954e4f4d271ca50e : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/839296d701f24921b89ff4987614aab6 : SUCCESS in 5m 48s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b43faf48a4ce47fa82836623d56cfeca : SUCCESS in 6m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f918a283cd3445f893e268ba2b5b33ac : FAILURE in 27m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/52613739f55641df80304fc8b79ae11b : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/629df6c8284847c59c3c0bb54e859047 : SUCCESS in 1h 38m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b2c1b8b32be6463cb5a3e9dd9c6d119b : SUCCESS in 1h 59m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9ea8cf45aee94157af4beb83f0932e72 : SUCCESS in 1h 53m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1aabb5f65d444b86915fb02d7c1be872 : SUCCESS in 1h 56m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/776c9566f8264ca7a661c233f789574a : SUCCESS in 36m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5c13eff1886244e1854cba39e5038783 : SUCCESS in 1h 14m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8b50004d96a24f3cbc741f0c36447edc : FAILURE in 1h 24m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1f403aecc3ba4732a10ead3dce236a47 : SUCCESS in 39m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f46e65ca6a5d4584aa2ec88cca307a9c : SUCCESS in 1h 28m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/86f02af1ab414e399fc597bf90e071dd : SUCCESS in 1h 17m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5b257640bb6a48969e2a63439d71dbea : SUCCESS in 1h 35m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4670a18002b04feb9c3761487f110abc : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":14},{"id":"7c8cd7dd2c23711a680d53019db31753f0f49d45","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-09 07:38:59.000000000","message":"Patch Set 14:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/205/.","accounts_in_message":[],"_revision_number":14},{"id":"c64dfe02092b5bb8968a48bd26573e2121cffe68","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-09 07:47:12.000000000","message":"Patch Set 14:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-847131-14/FC/ : SUCCESS in 2h 29m 5s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-847131-14/iSCSI/ : SUCCESS in 2h 4m 7s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":14},{"id":"46a252a9f55c530317f96c231ee755fb464c3469","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-09 07:52:10.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/31/847131/14/check/DellEMC_PowerFlex_3.6/1c1759c/DellEMC_PowerFlex_3.6/1680 : SUCCESS in 1h 55m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":14},{"id":"f1ae377c9f3fc7a252a3a9ed9b082aea8d30a9ca","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2024-07-09 07:59:44.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"588ab7c6db8240956b04f2b87fbefe416163c979","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-09 08:08:02.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/7c6f35a4fe4349408179dcc9ead2e046 : DISK_FULL in 46m 46s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/b660723bf2234a01bc6d73a68d812b56 : DISK_FULL in 29m 27s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"95d0aee2b5038cce26f569a500a5fadd95a0f7da","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-09 08:39:10.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/14/check/DellEMC_Unity_FC/dbe1faa/DellEMC_Unity_FC/2195 : SUCCESS in 1h 23m 02s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/14/check/DellEMC_Unity_iSCSI/ba8161d/DellEMC_Unity_iSCSI/2204 : SUCCESS in 2h 11m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"7e231d7afddf26b4c1c892329c75b8cb13612bdd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-09 09:48:24.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7a73abaadb2c47ad8e18cd0a08834712 : SUCCESS in 1h 09m 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/30b0373f394d40edbe326a9ab7b7fc30 : SUCCESS in 1h 17m 24s","accounts_in_message":[],"_revision_number":14},{"id":"467acce2bf9ef9efc23387414ae53e9274e5c1c5","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2024-07-09 12:35:10.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0e4a906668b6c8c111ba5e4eed9176861b7187bf","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-09 14:57:13.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/14/check/DellEMC_PowerStore_iSCSI/1eee0ac/DellEMC_PowerStore_iSCSI/1497 : SUCCESS in 1h 20m 05s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/31/847131/14/check/DellEMC_PowerStore_FC/b47a6dd/DellEMC_PowerStore_FC/1693 : SUCCESS in 1h 07m 11s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/31/847131/14/check/DellEMC_PowerStore_NFS/a3d46b8/DellEMC_PowerStore_NFS/1191 : FAILURE in 1h 13m 46s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/31/847131/14/check/DellEMC_PowerStore_NVMeTCP/d94b528/DellEMC_PowerStore_NVMeTCP/1044 : SUCCESS in 1h 52m 38s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/31/847131/14/check/DellEMC_PowerStore_NVMeFC/d817003/DellEMC_PowerStore_NVMeFC/7 : FAILURE in 2h 52m 16s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":14},{"id":"487a36caef3cf3dbafb72a146f58328215e8ef01","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-10 06:21:24.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"9e97b31f07139bbdd84031dc92ff976416acd7ed","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-11 19:00:34.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/14/20240712001855/NEC-V-FC : SUCCESS in 3h 32m 25s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/14/20240712001855/NEC-V-ISCSI : SUCCESS in 3h 41m 35s","accounts_in_message":[],"_revision_number":14},{"id":"f9a35d5a564e3fc5ff77fab0aca34e2ba6fcdc97","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-07-17 13:32:20.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d6288a937e95227504a3d15fea207d7ff09279e3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-07-17 13:35:34.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"7129b253d0498986b43954bd87fb64f9bc12e9fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-18 05:30:22.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"1b556585de273178080959bd0134a142fc6c291d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-18 05:30:38.000000000","message":"Patch Set 15:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/43/","accounts_in_message":[],"_revision_number":15},{"id":"202f54d0c7230cb33857157fdecdfb465ca65c5d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-18 05:30:44.000000000","message":"Patch Set 15:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/43/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":15},{"id":"15c02d3156bc81dcf98d1321780ca0af3a328f65","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-18 05:31:27.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"9694a86cf1ffa260eda90cbf83c0eb7e9a0cb02a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 05:42:29.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24cc28cc677844cf83d72465c112bfe0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2077cbcbc96d46e5b05f32fd6ddbe16e : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b09ae00b9ce54423abe81fb339de99df : SUCCESS in 11m 08s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"8baa74aca1c45fe8bd629989b9e5d4883d2f03f1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-18 05:57:22.000000000","message":"Patch Set 15:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/7c4222c0d3ae46e283f6dae8070c95b9\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/23440d267d8b48178bc9fb9731d8605e : POST_FAILURE in 20m 01s","accounts_in_message":[],"_revision_number":15},{"id":"8c4272c1eeddcd74c6156d0bd3be512e98b16f7e","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-18 06:16:18.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/847131 : SUCCESS in 40m 23s","accounts_in_message":[],"_revision_number":15},{"id":"4e5dd2412506beec903094e2ee4a9bbb27150c0c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-18 06:21:24.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847131 : SUCCESS in 41m 57s","accounts_in_message":[],"_revision_number":15},{"id":"6a15bec69081144c71788f8562b5634cf04ea845","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-18 06:25:17.000000000","message":"Patch Set 15:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/268/.","accounts_in_message":[],"_revision_number":15},{"id":"407d7f795aa89e551e7ef88bab8329819a195ed6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-18 06:31:34.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/847131 : SUCCESS in 59m 35s","accounts_in_message":[],"_revision_number":15},{"id":"63bab9cea002097d0187fe2565efa38a41d50309","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-18 06:45:46.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/aa51c174580c4207990913c62671f847 : SUCCESS in 1h 12m 59s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fdef759a39f4454f8ac4a95420f302f6 : SUCCESS in 1h 04m 49s","accounts_in_message":[],"_revision_number":15},{"id":"d2512ac3325b3edaa5580d53e91e699095d7abe3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-18 06:54:04.000000000","message":"Patch Set 15:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/752/.","accounts_in_message":[],"_revision_number":15},{"id":"b664cc6b530cbf5972616ef31d48dbb718782a5d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-18 07:22:52.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/15/check/DellEMC_Unity_FC/3fb8df5/DellEMC_Unity_FC/2255 : SUCCESS in 1h 34m 13s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/15/check/DellEMC_Unity_iSCSI/32f0ead/DellEMC_Unity_iSCSI/2262 : SUCCESS in 1h 51m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"e38e962d8f36aafb0622d55bd4ca63fbcc10cedb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-18 07:27:32.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/31/847131/15/check/DellEMC_PowerFlex_3.6/19526a5/DellEMC_PowerFlex_3.6/1743 : SUCCESS in 1h 56m 44s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/31/847131/15/check/DellEMC_PowerFlex_4.5/ad2d36a/DellEMC_PowerFlex_4.5/622 : SUCCESS in 1h 51m 03s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/31/847131/15/check/DellEMC_PowerFlex_4.x_NVMeTCP/e303502/DellEMC_PowerFlex_4.x_NVMeTCP/579 : FAILURE in 13m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"e91f45646fcee257e67e97b210eaaa0f93162afa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 07:44:16.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fa8a2040d0a4d16908aa43dac7a96a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/276a889d587e45b2adbc44161be4743d : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2ff70998ef34435bd106cb06fe6b167 : SUCCESS in 9m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d9c2c5c449e435bb20c1c4a1bf8622e : SUCCESS in 7m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aaa5088d287b4d3c8f3e002c7f14f5f5 : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7145b4b61114a719e0ecc8057bf1974 : SUCCESS in 10m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/5f73e250b6534c6389ba7159697c3a21 : SUCCESS in 46m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d7f64f9ec9f04478ad9066b75cad0e43 : SUCCESS in 1h 18m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8595be90dbc2472e9aaae8e8f880cd30 : SUCCESS in 46m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d629c43bcd5b41378c42e4e6e4a7e92d : SUCCESS in 12m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aecc1d77ff0f4371bb9b54ec8f864d28 : SUCCESS in 4m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/25cb38c1d2d740d4a8da98f49345e945 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c28a4e013cca4c8a86afa1c660ed1985 : SUCCESS in 5m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/be6267188eb1486d9206e99c7aa9ee29 : SUCCESS in 6m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8184b2048fea4e2dbacacca14ceb972c : FAILURE in 44m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8d765af7d6354788af9ceff07659bfd0 : SUCCESS in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/98ff7e88a78f4a6287567c3a39f2a83a : SUCCESS in 1h 43m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0f407edc84f841918df6eb415b1a3c98 : SUCCESS in 2h 12m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b8dcd840b728458a92b730ad9375ec60 : SUCCESS in 1h 39m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8ede81254b294dcb91495c6b95625989 : SUCCESS in 2h 04m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5c6ea29f0fce4b8ab3bcccae0d830d04 : SUCCESS in 43m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ce1245c47161437eb6e372af815c5fb7 : SUCCESS in 1h 32m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1ee0989fdcbe446e8243dfc6a6ed8153 : SUCCESS in 1h 27m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/29c27b7226a24105932fedb458a261b5 : SUCCESS in 1h 02m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a970a02a510e4ee089173e84ad0c7382 : SUCCESS in 1h 41m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0caa937543834a32bdcdf4368ca588e5 : POST_FAILURE in 1h 47m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b59376a871e54291bed5bd080297992f : SUCCESS in 1h 10m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/682650d817e34abd86e2fbcda054ecb7 : SUCCESS in 37m 57s","accounts_in_message":[],"_revision_number":15},{"id":"47525bf1ae4ed3e8faa8afa5c0711fe0843e29e3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-18 07:50:04.000000000","message":"Patch Set 15:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_07_2024_07_44_26\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":15},{"id":"14e10a2cc68a42f3defb880e5effe8f1fcb900f5","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-07-18 08:00:08.000000000","message":"Patch Set 15:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/847131/15/fc : SUCCESS in 0h 45m 22s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/847131/15/iscsi : SUCCESS in 0h 48m 58s","accounts_in_message":[],"_revision_number":15},{"id":"e2805668887ff38cfc275cac585b82d80fe42397","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-18 08:08:13.000000000","message":"Patch Set 15:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-847131-15/FC/ : SUCCESS in 2h 37m 24s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-847131-15/iSCSI/ : SUCCESS in 1h 51m 59s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":15},{"id":"27b28f4dff15594af35031d313898663cd7f7282","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-18 08:15:30.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/15/upstream-check/cinder-tempest-plugin-ontap-iscsi/5da516f/ : RETRY_LIMIT in 9m 10s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/15/upstream-check/cinder-tempest-plugin-ontap-nfs/2ee4491/ : RETRY_LIMIT in 8m 24s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"940f9771f86859469ed464ad9bcb4e07e14d3c7c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-18 09:33:26.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/15/check/DellEMC_Unity_FC/4761bb7/DellEMC_Unity_FC/2256 : SUCCESS in 1h 27m 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/15/check/DellEMC_Unity_iSCSI/407bb75/DellEMC_Unity_iSCSI/2263 : SUCCESS in 1h 48m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"977cdf03c8dfdfbca5ed20fa234a73e4673debc3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-18 09:42:45.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/31/847131/15/check/DellEMC_PowerFlex_3.6/d038e5a/DellEMC_PowerFlex_3.6/1744 : SUCCESS in 1h 48m 05s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/31/847131/15/check/DellEMC_PowerFlex_4.5/7e4b7fa/DellEMC_PowerFlex_4.5/623 : SUCCESS in 1h 58m 11s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/31/847131/15/check/DellEMC_PowerFlex_4.x_NVMeTCP/e856ef0/DellEMC_PowerFlex_4.x_NVMeTCP/580 : FAILURE in 13m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"6e0e7d68d5851f99f0d2d1542b6958ae5536d952","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-18 10:36:53.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/15/check/DellEMC_PowerStore_iSCSI/40f857e/DellEMC_PowerStore_iSCSI/1523 : FAILURE in 1h 21m 43s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/31/847131/15/check/DellEMC_PowerStore_FC/ef592b5/DellEMC_PowerStore_FC/1720 : FAILURE in 1h 11m 34s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/31/847131/15/check/DellEMC_PowerStore_NFS/17b2f9b/DellEMC_PowerStore_NFS/1218 : FAILURE in 1h 07m 49s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/31/847131/15/check/DellEMC_PowerStore_NVMeTCP/ed71213/DellEMC_PowerStore_NVMeTCP/1070 : FAILURE in 1h 57m 58s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/31/847131/15/check/DellEMC_PowerStore_NVMeFC/0812725/DellEMC_PowerStore_NVMeFC/37 : FAILURE in 2h 52m 16s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":15},{"id":"fc6bbe8dcaf14fbf04c2ba535b37af1605c67ca0","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-18 13:43:38.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/15/20240718184005/NEC-V-FC : SUCCESS in 3h 53m 31s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/15/20240718184005/NEC-V-ISCSI : SUCCESS in 4h 03m 30s","accounts_in_message":[],"_revision_number":15},{"id":"b36c5eac054fa52820264400be5a06f386fa4f0d","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2024-07-18 16:27:01.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"acfccb5189cd38da9935c6b0cd39dda9fe2854b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-20 15:56:08.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"ed76d5d73c60ef0f677c51372433140144de769f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-20 15:56:15.000000000","message":"Patch Set 16:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/102/","accounts_in_message":[],"_revision_number":16},{"id":"57f0883f593f96dc90e289f7fa1e34bf0d8ee07a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-20 16:07:02.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc3d4d5a080a4d2299b4ddbe0844c797\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cad5bcdc998b40d9ad58ea400c22f462 : SUCCESS in 9m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a0f36f85c85a494cae8ce88ce1405c9e : SUCCESS in 9m 42s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"00d9b0c15377eb75951723500274382785cff794","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-20 16:17:22.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"c59f0f406f8ee55572a84845ea1d072cbac64503","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-20 16:18:11.000000000","message":"Patch Set 16:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/102/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":16},{"id":"500827602057a7107135d3705533281ef046dbfd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-20 16:27:27.000000000","message":"Patch Set 16:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/7e84c5452b9c4309bf7631d4fc0441c9\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/880e323d9276403c8a1bcbdbb4a1e36b : POST_FAILURE in 19m 23s","accounts_in_message":[],"_revision_number":16},{"id":"bb409bbaa49d275baf709ea836caac2e272e8ef4","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-20 16:36:57.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/847131 : SUCCESS in 39m 31s","accounts_in_message":[],"_revision_number":16},{"id":"b9262d66683fb3c23fd557092c75df47459bbe77","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-20 16:47:14.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847131 : SUCCESS in 42m 01s","accounts_in_message":[],"_revision_number":16},{"id":"c59727e6af98fc70f909c4fca80ffce8c4fed380","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-20 16:56:41.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/847131 : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":16},{"id":"6bdff5f31ddfc71af4ea5ff3ceb110a80fa71e20","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-20 17:17:59.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/adbd8de49e8a44f1ad136eaff1996154 : SUCCESS in 1h 19m 49s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6c44f11858fe4e388f72c56d6923cb1b : SUCCESS in 1h 07m 04s","accounts_in_message":[],"_revision_number":16},{"id":"6ca4600c723a1d0abf97bcf324dcd38d081fe545","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-20 17:39:29.000000000","message":"Patch Set 16:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/765/.","accounts_in_message":[],"_revision_number":16},{"id":"0c7804296da6a9532159f17ea1e81af29fd962d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-20 17:50:54.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ac87cebbe5b45eb89b504826d2b918d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aed545e01ad4a4db7b0aad072bb9f08 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd851e6d065844d985b25db7db02856d : SUCCESS in 12m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4f5492f00824d65ab2ea4a1b02404f7 : SUCCESS in 10m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/588c804725d843369ce8fdfa397a13d9 : SUCCESS in 14m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4faa20513ba74af387fa914f3b43d332 : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/7438a4156e5c4cfdad3e942eb18a8da7 : SUCCESS in 40m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c217797adbfc4ff1889a3ef9b40d3c6b : SUCCESS in 1h 26m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f5fc5cfb6344cda9bdc50eebd604e54 : SUCCESS in 41m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3822d29deee24eecbf0213a69314e55a : FAILURE in 3m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ae2757e4ef764d78854b3c0e4e34b34a : SUCCESS in 12m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4e831ede022d48b0bd6063d4ed0ee7ec : FAILURE in 4m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9f2d8f0971c04c5685709d748602fd8b : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aa04bd7f8a0c423bb970f4c90f7adb73 : SUCCESS in 5m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6257470ab2dd4cae964a0766e4ff07da : SUCCESS in 5m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/66867c3ad1694550b3b8b1e6cb86fddd : FAILURE in 23m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5ed62e1b25b0455595610f74a4049624 : SUCCESS in 6m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e06ba57e3d924ba3ba964454b148827b : SUCCESS in 1h 22m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c27f450887a64b52b69aa322df87a52f : SUCCESS in 1h 52m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/07b8dcb194ba4d41834867c822f6809f : SUCCESS in 1h 23m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3eecb849bf7b479c8febd438be6c2943 : SUCCESS in 1h 37m 38s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6e42d7d2964449f09fcc2a475c0d1c2d : SUCCESS in 21m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/13ed6e36b84440c7be445fb6d26783fb : SUCCESS in 51m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8ba598f383204718a1ec96472b7b4787 : FAILURE in 1h 22m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/75277343a30a4fe6bdf2a19dc867b577 : SUCCESS in 30m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4b68409555de40469d98e27e2d9d96eb : SUCCESS in 1h 22m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e8105ffc50294c60a22e33aac0046e4b : POST_FAILURE in 1h 53m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dec02a72e10e4664a713f275ca734c50 : SUCCESS in 1h 36m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a50677a707b1457e8c51740dd30f473f : SUCCESS in 55m 48s","accounts_in_message":[],"_revision_number":16},{"id":"145b7623cdf2796b188384af053f5b3a50c5eb04","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-20 17:53:28.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/16/check/DellEMC_Unity_FC/2245cfe/DellEMC_Unity_FC/2269 : SUCCESS in 1h 35m 56s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/16/check/DellEMC_Unity_iSCSI/67b84c5/DellEMC_Unity_iSCSI/2276 : SUCCESS in 1h 56m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":16},{"id":"aeab8ac4909a72b2c099c51c404156f1276ec9eb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-20 18:00:27.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/31/847131/16/check/DellEMC_PowerFlex_3.6/b6c33eb/DellEMC_PowerFlex_3.6/1757 : FAILURE in 2h 03m 57s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/31/847131/16/check/DellEMC_PowerFlex_4.5/29d64e8/DellEMC_PowerFlex_4.5/636 : SUCCESS in 1h 49m 32s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/31/847131/16/check/DellEMC_PowerFlex_4.x_NVMeTCP/b940f4a/DellEMC_PowerFlex_4.x_NVMeTCP/594 : FAILURE in 13m 07s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":16},{"id":"55787c000a91a3370ca3e0f09ad0e0bfc82203df","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-20 18:40:23.000000000","message":"Patch Set 16:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-847131-16/FC/ : SUCCESS in 2h 43m 56s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-847131-16/iSCSI/ : SUCCESS in 2h 22s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":16},{"id":"9143c060f8e5bec2450369ab79cbdbedd07c25e4","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-21 06:16:46.000000000","message":"Patch Set 16:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/16/20240721060946/NEC-V-FC : FAILURE in 8h 30m 32s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/16/20240721060946/NEC-V-ISCSI : FAILURE in 9h 06m 56s","accounts_in_message":[],"_revision_number":16},{"id":"eef9698454fa9dec34c6fc7b8f2ab68cdef01b4f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-21 13:49:27.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/16/20240721170326/NEC-ISCSI : SUCCESS in 5h 18m 13s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/16/20240721170326/NEC-FC : SUCCESS in 5h 45m 57s","accounts_in_message":[],"_revision_number":16},{"id":"e3bb3638600e4e6d78c7dac0ba9b7d5500333206","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-23 04:46:03.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"a37105b13e37afd02b1dd8e719cf8145e3194410","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-23 04:46:12.000000000","message":"Patch Set 17:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/117/","accounts_in_message":[],"_revision_number":17},{"id":"fcf8cee62dd3c0f8d2b80e01c7999bbe81b6576a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-23 04:46:15.000000000","message":"Patch Set 17:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/117/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":17},{"id":"5d752ebad4ac1f4aa888e42292305d28f86afde5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 04:58:23.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db247f42e855425c997fa27ba9220b1c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3fe293e2ab6f47f8aa7cce5ce9938e88 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ce90af3666c949a4ba726951705c8dc1 : SUCCESS in 11m 24s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"4c0a4598e79620aa0aad794a87366f81050b580c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-23 05:15:18.000000000","message":"Patch Set 17:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1b4649cbfb8043b1a19e95f0827ed5ff\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f8384deeefef495785caf01acab8d51d : POST_FAILURE in 20m 06s","accounts_in_message":[],"_revision_number":17},{"id":"1aa375f9b1ebd51057815c57a21955a92b59a82a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-23 05:37:06.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847131 : SUCCESS in 41m 51s","accounts_in_message":[],"_revision_number":17},{"id":"50ed62e02a32053ec253a5d9f3a812f464cdd7db","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-23 05:47:29.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/847131 : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":17},{"id":"fabf8eb32672de75f86c6644ea69c43e465eb35c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-23 06:05:24.000000000","message":"Patch Set 17:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/776/.","accounts_in_message":[],"_revision_number":17},{"id":"97cd62d4cf3f34f6fd4f77813abd14c5535f2001","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-23 07:03:19.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/60e9d1c7056841c3b9abff1f84c47392 : FAILURE in 1h 15m 53s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a88c480d48184e669442501fa1e58bb5 : SUCCESS in 1h 08m 27s","accounts_in_message":[],"_revision_number":17},{"id":"7951bdf8db3e4fcdee913e495761771ef4d2c0ab","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-23 07:04:35.000000000","message":"Patch Set 17:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/17/check/DellEMC_Unity_FC/298064a/DellEMC_Unity_FC/2280 : SUCCESS in 1h 36m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/17/check/DellEMC_Unity_iSCSI/8bf1344/DellEMC_Unity_iSCSI/2287 : SUCCESS in 1h 51m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":17},{"id":"0e8c8c957fdb9c364ef22b154605a9b227164ea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 07:14:48.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/06200af71f2a474896af144c5e4687fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d49f28d2fc24f95a1a39a87b88dbb1c : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d5a0fb9d6e64930a81af96017aa177d : SUCCESS in 8m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/72bda632d0a1428a829f70e67914507d : SUCCESS in 9m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ee1c29bf16840abaf6167607c9880f5 : SUCCESS in 11m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c29256eef9cb466bad88fe4e6032f44c : SUCCESS in 11m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/e4560c88026144b19779ee8ddbc4384e : SUCCESS in 53m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/08a9b8feddd74ed78c4ca0f7556a574e : SUCCESS in 1h 30m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e8aece316ba4beeb2763447de3aed2d : SUCCESS in 1h 02m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af893e50affd462491fc1c462216df22 : FAILURE in 3m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6663e74e09e1464681a015ce2dba7bbd : FAILURE in 15m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/37fe482893834a349b1c54050fd7d664 : SUCCESS in 4m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2890ac31729143e7a79c253c2d98f9a4 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f4b56e2630764a5c9b4683f86972daff : SUCCESS in 4m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/66cc19098b344ead835f1e22170a91f9 : SUCCESS in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e0a5676c81e84651b603fce0c8c922f9 : FAILURE in 30m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/03bfe50a739942b0bedc47b7988e358e : SUCCESS in 4m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c5258b6b51ed4c9687ce706da2b69628 : SUCCESS in 1h 25m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/be57e8f664e74d56960375aa61b85bf6 : FAILURE in 1h 53m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d31518badd2c4db1b243e2f8c21c18cf : SUCCESS in 1h 41m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/11517c64f1c943469819fc321c03a612 : SUCCESS in 2h 21m 29s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9077325514cf482385b78cd353d307cd : SUCCESS in 23m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2e27b8d4d0b4470681f7a916ece25780 : SUCCESS in 1h 11m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b262ff47a77c47ec8a58c5fe8191b8af : SUCCESS in 1h 38m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f834b5b129054696bf4bc2e9b0987d14 : SUCCESS in 25m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cecdd59ded8547a3bd250129ca303dc3 : SUCCESS in 1h 06m 49s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b1e6ea21d11b44b2abf04c14d38b7181 : SUCCESS in 1h 48m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e51e6ee7a6274ba58bdd36e216c5d230 : SUCCESS in 1h 40m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1caeb293a3284632b3ff94f98b50b2f9 : SUCCESS in 58m 43s","accounts_in_message":[],"_revision_number":17},{"id":"f9920ff1512b9f268bffc7672527c7efdadc8696","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-23 07:19:48.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/31/847131/17/check/DellEMC_PowerFlex_3.6/b2c5989/DellEMC_PowerFlex_3.6/1768 : SUCCESS in 1h 52m 03s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/31/847131/17/check/DellEMC_PowerFlex_4.5/8ee05b1/DellEMC_PowerFlex_4.5/647 : SUCCESS in 1h 43m 52s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/31/847131/17/check/DellEMC_PowerFlex_4.x_NVMeTCP/e9508db/DellEMC_PowerFlex_4.x_NVMeTCP/605 : FAILURE in 13m 36s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":17},{"id":"a8d7341257f56deb314f6572ca235776a138d1a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-23 07:25:23.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"7b6ecded2a627df90b85ac85259fbfa5eafc91fe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-23 07:25:33.000000000","message":"Patch Set 18:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/118/","accounts_in_message":[],"_revision_number":18},{"id":"d359f11b3eeaff0c29b269a3c093ce67e658588b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-23 07:25:35.000000000","message":"Patch Set 18:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/118/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":18},{"id":"0bf457ee488e210f8b2eb8788a07fa36755f5372","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 07:34:54.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba0172c2dfac44d19abf02652dd7db1c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7aa5a98ff3e8417884132883d4422d8f : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5415fe64448b4686a75f15182d07c6df : SUCCESS in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"f7eefb5372fe74dfab0a8554ed3bec76a3dcdcfb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-23 07:50:07.000000000","message":"Patch Set 18:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5810e41abe024db5b45f2339f22b9cd1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c9b3e8f4a9674f5baaaa298b934b8c0a : POST_FAILURE in 20m 03s","accounts_in_message":[],"_revision_number":18},{"id":"238da498467a202561216f0c1c7a3bf181eb59bf","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-23 08:10:27.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/847131 : SUCCESS in 43m 20s","accounts_in_message":[],"_revision_number":18},{"id":"d3306a103fc0f80cad67f249822149312e91557f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-23 08:21:50.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847131 : SUCCESS in 43m 54s","accounts_in_message":[],"_revision_number":18},{"id":"13e9f391f0caefea9d55431c54c7cfd3b6c5fbd9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-23 08:32:00.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/847131 : SUCCESS in 1h 05m 13s","accounts_in_message":[],"_revision_number":18},{"id":"867b4c973070e5963b93accf799d1991bbb0e2fc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-23 08:45:46.000000000","message":"Patch Set 18:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/777/.","accounts_in_message":[],"_revision_number":18},{"id":"f49eb6bd48a058e4ba327e563110bdc5316db110","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-23 09:43:24.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e5bf89b6a2674547a971a124888e05e8 : FAILURE in 1h 13m 35s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1454a2cd942b445caf60e97742e5d8d7 : SUCCESS in 1h 08m 57s","accounts_in_message":[],"_revision_number":18},{"id":"4f72af49a778347b744d66e82f5e1664a1b06652","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 09:46:09.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a049bb6e6cc4400ea7650e3e2b85932a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68da7017ab58410fbf7da3db3619f434 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5db55bb8cde414fbb8af73f445a1e75 : SUCCESS in 10m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5669f9ddc794e6ead5c42d75fbd1a89 : SUCCESS in 8m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85d6b10d1e70462d95c7b5d1e17feb02 : SUCCESS in 14m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df05c6dbee14e62a73c76e920ef7fc6 : SUCCESS in 13m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/085d622d8bb64ca6bac988aaf29d1a30 : FAILURE in 28m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9cace91f3d404c759d31d0b1a2c32fc9 : SUCCESS in 1h 33m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94993649ec094cb2964633887e8dfec8 : SUCCESS in 1h 01m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb623a88895143fe9ebc5a42be341eb0 : FAILURE in 3m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8939b319cc2a45f4a91d9e46281a77fc : SUCCESS in 16m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0d261a723021489f87a874cf102f734e : SUCCESS in 4m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1dcb202b327444b3a187011177b150c7 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/24e98cc4a795467b845805dd30c616a6 : SUCCESS in 8m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/80dfb77d1bb248698a23eb7a77efce5c : SUCCESS in 5m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/59c72e94e77d4be28771dfdf1b9f8bec : FAILURE in 47m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9ff331a769b94bcea10d8007d461c7b3 : SUCCESS in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ebf5f5c15c154c7597be25bed18c8cda : SUCCESS in 1h 46m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9d190cbac3a24e0fa48e083eb3591de3 : SUCCESS in 2h 00m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/72b8e7568a21491295228bd19ed28d44 : SUCCESS in 1h 35m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2cd3af8b183549cbb04ca16c66ce5a8d : SUCCESS in 2h 00m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8c9f9e78411a43de92c109c216ec38c3 : SUCCESS in 37m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1f855023b59741aba2f8c1c06eb43f00 : SUCCESS in 1h 15m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6f23bc67ff7847ec9dff5434ccbecdf8 : SUCCESS in 1h 41m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e56deb4753374384b5b6293bc54515f8 : SUCCESS in 44m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/adabdb5079af413e94628315dce2c022 : SUCCESS in 1h 01m 43s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4f35f001852d42e99d98ad47468194ea : SUCCESS in 1h 13m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f4e660d0721849fd8c8b8333bdaa3e71 : SUCCESS in 1h 47m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f5c46c6b18949cc8be8ab90e27126eb : SUCCESS in 41m 39s","accounts_in_message":[],"_revision_number":18},{"id":"a7f8263483023e8219b0c15484c16fee310362cf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-23 10:40:14.000000000","message":"Patch Set 18:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/18/check/DellEMC_Unity_FC/8ad27a2/DellEMC_Unity_FC/2282 : SUCCESS in 1h 26m 24s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/18/check/DellEMC_Unity_iSCSI/e731014/DellEMC_Unity_iSCSI/2289 : SUCCESS in 1h 48m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":18},{"id":"5f4e7dabced2bf64b98e8ee10b06fd30e7376a68","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-23 11:05:27.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/31/847131/18/check/DellEMC_PowerFlex_3.6/b8f14e1/DellEMC_PowerFlex_3.6/1770 : SUCCESS in 1h 52m 48s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/31/847131/18/check/DellEMC_PowerFlex_4.5/76e3125/DellEMC_PowerFlex_4.5/649 : SUCCESS in 2h 00m 07s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/31/847131/18/check/DellEMC_PowerFlex_4.x_NVMeTCP/842a389/DellEMC_PowerFlex_4.x_NVMeTCP/607 : FAILURE in 14m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":18},{"id":"44aa1878c305c5871f24c8ee6584b2fad26ee70e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-23 11:25:56.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"deabf3a3ffcbd721ec5cc6e5357b6b360de5c2aa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-23 11:26:08.000000000","message":"Patch Set 19:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/120/","accounts_in_message":[],"_revision_number":19},{"id":"68ee8f8f1b8b65e0fd481b2821b954bc8b0f386d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-23 11:26:10.000000000","message":"Patch Set 19:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/120/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":19},{"id":"2eb1022d2163a87a6b60e98f7d2ce25c80d3531f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 11:51:32.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9415e22d830245af84a502001bfae6d9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2a1b967eb954eb0b9304c1374323fcd : FAILURE in 20m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9b506e5a429c4b6fb6ae73580b29cbfe : SUCCESS in 7m 32s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"cc9c6724ef68c57a749b44d75d11e649c8b213f2","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-23 12:08:16.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/847131 : SUCCESS in 40m 50s","accounts_in_message":[],"_revision_number":19},{"id":"2987eb4408e0433608550965077a22068a6648e8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-23 12:19:38.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847131 : SUCCESS in 44m 43s","accounts_in_message":[],"_revision_number":19},{"id":"0dfbe168034707fa00bd86beb63570cc32b187fb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-23 12:27:20.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/847131 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":19},{"id":"66f2f075cbfd98cc9ef1133552874423bca28fc3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-23 13:29:07.000000000","message":"Patch Set 19:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/780/.","accounts_in_message":[],"_revision_number":19},{"id":"b3c60a175364030d83275ad49544b83e4b48843a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 14:24:22.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b58c55fd98249ecb231a8ff941fa915\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/625a6eae77024e508c27b2358bb75f53 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a69fc93f13e249caa972114ad4bfdac0 : SUCCESS in 10m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ed5532ee1e94eb2b56f0db00810ba80 : SUCCESS in 11m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c5ba94ffdd6f4a0b9082291afc15626d : SUCCESS in 16m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5269dff535e4a55942f0ef658fc8f37 : SUCCESS in 13m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/5e21b6a115454a3ea367e583da860ac0 : SUCCESS in 1h 17m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/acb2111475c44b2c87312fa1a8878e06 : SUCCESS in 1h 40m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7656db0a58614b4698ebd0d6a0b2cad9 : SUCCESS in 55m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf1cf9faebe24362ac21963612c3e54e : SUCCESS in 20m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/18c0f031efd440ef97502255b684ba49 : SUCCESS in 17m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3886c553799748d78a6a721fac4c10f6 : SUCCESS in 6m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eb3ac77f2f6c45328180601d9627fb21 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ba0e8c34de94a76969a629e0dd634b9 : SUCCESS in 5m 51s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cd2d94a5398f432780d883d268ac9028 : SUCCESS in 7m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/66a796b691254b568ca0d9d617e37474 : FAILURE in 33m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7a4faf0cd3b843959e5dfd07632827e7 : SUCCESS in 5m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cc153b953464464694c3fa4cd34371dd : SUCCESS in 1h 39m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2f6005b4557c4c88a5c45db77f32e0f7 : SUCCESS in 2h 47m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/39dccca32e8643a0830a9b45dafc7371 : SUCCESS in 1h 49m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/a7362c9ed7634ed2b6c74624ddac0202 : SUCCESS in 2h 20m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b87bb38dad894543912dc1e8708c1ca1 : SUCCESS in 25m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2b63822b84a04c588cff827043c4cdc1 : SUCCESS in 1h 06m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/48899916efff4c34991767b040dea048 : SUCCESS in 1h 20m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8549cdd7e1f140f795431f30ce565ec8 : SUCCESS in 38m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dfd00789e31f49908140e087da0a8761 : SUCCESS in 1h 21m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ed6d4b0210af468990463baadc459399 : SUCCESS in 1h 29m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bca7f0807d2d4af19a5913c8359983e5 : SUCCESS in 1h 33m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bdf1b8a81b4742a58d1c53af37d652fe : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":19},{"id":"a6b5a2b8c46bd8ceb2e0f919b1d98d4cf629ba8c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-23 14:29:23.000000000","message":"Patch Set 19:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_07_2024_14_24_33\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":19},{"id":"003d1c6c8910a0fa99f93ac02342f93f7ffe4f91","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-23 14:33:18.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/19/upstream-check/cinder-tempest-plugin-ontap-iscsi/d106227/ : RETRY_LIMIT in 4m 09s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/19/upstream-check/cinder-tempest-plugin-ontap-nfs/6bec91a/ : RETRY_LIMIT in 8m 14s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":19},{"id":"286f65b7f8931f2fcb401db10c3c938e387f8057","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-23 14:56:02.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/809a802a157243efb72785c6bdd4b75e : SUCCESS in 1h 18m 57s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c763670d22054717bc0148dfc5e8fa6c : RETRY_LIMIT in 2h 46m 24s","accounts_in_message":[],"_revision_number":19},{"id":"62f0fd6bb6dd58b618deac5f37e41d28898d08c6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-23 15:05:52.000000000","message":"Patch Set 19:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/a8b028bcdb0c488092e09954ca298135\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7733300459644d569f371823cfcafdba : POST_FAILURE in 25m 59s","accounts_in_message":[],"_revision_number":19},{"id":"16e1a300180d74547589fb34d4b5986d4fb75b5f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-23 16:43:01.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/19/check/DellEMC_Unity_FC/e9fb94b/DellEMC_Unity_FC/2284 : SUCCESS in 1h 25m 34s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/19/check/DellEMC_Unity_iSCSI/ed4d643/DellEMC_Unity_iSCSI/2291 : SUCCESS in 1h 57m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":19},{"id":"bb725627024de56121ad3c5bc1ae2ccfda80b205","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-23 16:43:12.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/31/847131/19/check/DellEMC_PowerFlex_3.6/6e5e1d7/DellEMC_PowerFlex_3.6/1772 : SUCCESS in 1h 48m 27s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/31/847131/19/check/DellEMC_PowerFlex_4.5/05683f9/DellEMC_PowerFlex_4.5/651 : SUCCESS in 1h 55m 19s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/31/847131/19/check/DellEMC_PowerFlex_4.x_NVMeTCP/77b6b20/DellEMC_PowerFlex_4.x_NVMeTCP/609 : FAILURE in 13m 21s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":19},{"id":"9baf526b35b7140f0d864c45be8072da1930f2a6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-23 17:53:59.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/19/check/DellEMC_PowerStore_iSCSI/14baaaf/DellEMC_PowerStore_iSCSI/1537 : SUCCESS in 1h 36m 54s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/31/847131/19/check/DellEMC_PowerStore_FC/d16e27e/DellEMC_PowerStore_FC/1733 : SUCCESS in 1h 01m 25s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/31/847131/19/check/DellEMC_PowerStore_NFS/092358c/DellEMC_PowerStore_NFS/1231 : FAILURE in 1h 17m 29s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/31/847131/19/check/DellEMC_PowerStore_NVMeTCP/694eb98/DellEMC_PowerStore_NVMeTCP/1090 : SUCCESS in 2h 00m 12s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/31/847131/19/check/DellEMC_PowerStore_NVMeFC/81b7837/DellEMC_PowerStore_NVMeFC/54 : SUCCESS in 1h 14m 37s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":19},{"id":"b91765207ac830914f80652a71e4dcb9f760bc74","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2024-07-23 19:00:03.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"31548bb5d700738eed3afae413b8f4e04b01c5ea","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-07-23 19:00:08.000000000","message":"Patch Set 19:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/847131/19/fc : SUCCESS in 0h 41m 13s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/847131/19/iscsi : SUCCESS in 0h 57m 22s","accounts_in_message":[],"_revision_number":19},{"id":"7388ad5cf226042be4c9b9f3ede999c1508ee8ee","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-23 22:05:42.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a0c89109bd61473dbe5d574053377aec : SUCCESS in 1h 14m 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8b6eceea55c24a6ab6f8b0077f16d151 : SUCCESS in 1h 08m 53s","accounts_in_message":[],"_revision_number":19},{"id":"2703ec04ad9b2a01e6410e5fb1c7c0e9cb305b30","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-23 23:40:17.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/19/20240724044836/NEC-V-FC : SUCCESS in 3h 49m 50s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/19/20240724044836/NEC-V-ISCSI : SUCCESS in 3h 51m 37s","accounts_in_message":[],"_revision_number":19},{"id":"c14e90a4bdf79b1491d7433ab734b6ae4aefbf76","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-07-24 06:29:46.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"49085ebda00705521adc1dcb83e7444294feb261","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-24 06:42:54.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"5ce520fbdd3dfd24ffebdf4d89669523be710656","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-24 06:43:05.000000000","message":"Patch Set 20:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/124/","accounts_in_message":[],"_revision_number":20},{"id":"6689c4882ed7fa01e8e9e3c6c86a92bd3801a6b2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-24 06:43:07.000000000","message":"Patch Set 20:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/124/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":20},{"id":"03a368efaeb6ed9971cd238a635f401e4063e1e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-24 06:44:29.000000000","message":"Uploaded patch set 21: Commit message was updated.","accounts_in_message":[],"_revision_number":21},{"id":"43d752ccbfde96d0839efff6e9e447d1f6293cb8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-24 06:44:40.000000000","message":"Patch Set 21:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/125/","accounts_in_message":[],"_revision_number":21},{"id":"89139381686277a830e64dd252515ac15c776e50","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-24 06:44:42.000000000","message":"Patch Set 21:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/125/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":21},{"id":"01fc3d4c397d232c556c2b3bc93f69eb9ee4f1ae","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-24 06:44:45.000000000","message":"Patch Set 21:\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"2f47ec2ec3f77a2a0cef2daceae608fa44bd2d93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 06:57:28.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f46fac5426584350b7b3f31e9c7c0207\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a5430161bc640159b28595ecc9d6a1c : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/38bdd4202c3d4e0e8947215ca3a3667f : SUCCESS in 10m 34s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"59f79242b882f43bb318125411a4202f29052982","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-24 07:11:19.000000000","message":"Patch Set 21:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/7170376f55c24b9880766c9c6ab86552\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/4c852f31079f46cab98412665260cb77 : POST_FAILURE in 24m 12s","accounts_in_message":[],"_revision_number":21},{"id":"a3890d6a5279b43dcaeb94c026dc6bcdbfcdff08","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-24 07:24:53.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/847131 : FAILURE in 36m 10s","accounts_in_message":[],"_revision_number":21},{"id":"a87600ba2b952c77ff2a26a1a4f1801b58f02dd3","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-24 07:33:15.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/847131 : SUCCESS in 47m 02s","accounts_in_message":[],"_revision_number":21},{"id":"ee4875d528462b6e7111c31bc4cced9149e11135","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-24 07:46:27.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847131 : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":21},{"id":"7a859532730e7e680cb5d18e612fb170a0adcd46","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-24 08:08:31.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ce8ca6db5959491fa279d567cb2b9f3b : SUCCESS in 1h 22m 48s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1b055044811c4496a49dc018624824ee : SUCCESS in 1h 10m 44s","accounts_in_message":[],"_revision_number":21},{"id":"2d05c4a10f9e3bdc33120296bf4740569e2de00f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-24 08:37:01.000000000","message":"Patch Set 21:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/21/check/DellEMC_Unity_FC/4997758/DellEMC_Unity_FC/2289 : SUCCESS in 1h 24m 37s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/21/check/DellEMC_Unity_iSCSI/075a2ed/DellEMC_Unity_iSCSI/2296 : SUCCESS in 1h 52m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":21},{"id":"ce456ca649b582b861cd293d4779ec3ae9a8f32a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-24 08:56:08.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/31/847131/21/check/DellEMC_PowerFlex_3.6/aae279a/DellEMC_PowerFlex_3.6/1777 : SUCCESS in 1h 59m 19s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/31/847131/21/check/DellEMC_PowerFlex_4.5/f59c3c7/DellEMC_PowerFlex_4.5/656 : SUCCESS in 2h 11m 20s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/31/847131/21/check/DellEMC_PowerFlex_4.x_NVMeTCP/d44ea8e/DellEMC_PowerFlex_4.x_NVMeTCP/614 : FAILURE in 12m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":21},{"id":"8d284b5feda12ec80d6657184978e093fcb6d306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 09:08:59.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fbb3636c10e4e77aef5be713f145957\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/def5a4c242af49c485fc912c9b058544 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0fd3ca72db34213a445a31ff2c7f349 : SUCCESS in 12m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/215e143e776a4bd5aef37e1c3cd3b7f0 : SUCCESS in 10m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6a36753ed7e747d1aad7c79bca6fdb5b : SUCCESS in 15m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c76a5437d90a4ae7b8874a70bde68514 : SUCCESS in 10m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/bc7589ab5e834d0db02d9cb2a5ac056b : SUCCESS in 1h 00m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b10ff00137974d9d90e0d8dc9e45fd83 : SUCCESS in 1h 28m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17b36e89887346828656a23cd8493ae2 : SUCCESS in 59m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb0c971630364e528c87bbab16fda4f7 : SUCCESS in 19m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/66e5d3190d634e07bfda9f3123d11da6 : SUCCESS in 16m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/64f0d420420047b79c565bb646020efb : SUCCESS in 5m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/01a714835be5414ea85bd03a08f13e3b : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/568ea009cad94d1db444c4928cbea6da : SUCCESS in 5m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4fd7a82b722145f3aee54be11b5f1cfe : SUCCESS in 5m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/12f954eecd5f44e491ea5ba731c61571 : FAILURE in 25m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1686c86508f14e6a8fa35988884c9cea : SUCCESS in 4m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cc4ca2bdc3014527a5c9d782aea319d7 : SUCCESS in 1h 26m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ec548d3e191b43a1ae9bf8a6b5e8a1c8 : SUCCESS in 2h 18m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cdbfb89c6b4c46ae836575c39c782398 : SUCCESS in 1h 37m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/642c594b07dd4927950e8448c3b2ee80 : SUCCESS in 1h 52m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cf2f80b5a5534d6d973620d3946b144d : SUCCESS in 22m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8fe037772bdf4a06bb729cdcdd452108 : SUCCESS in 1h 04m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/064340a6608541d186e11b0e59826e4f : SUCCESS in 1h 13m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/da548ab43f604ec2ac6f46ec1be340d7 : SUCCESS in 39m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cbdd5e2c1e5a46008a92d5c503a19de7 : SUCCESS in 1h 22m 34s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2cc9f866b3094e188898887c0d21652b : SUCCESS in 1h 39m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ee68055d1dea41f0adf629bb0e44e41a : SUCCESS in 1h 39m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f2fea65be914244978fc77b4ea04982 : SUCCESS in 43m 19s","accounts_in_message":[],"_revision_number":21},{"id":"230b83a173beb1981f1f1c8b88ae5d5f89d440ed","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-24 09:14:49.000000000","message":"Patch Set 21:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_07_2024_09_09_12\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":21},{"id":"2a4f11a15f340795f31a22c61f77ec93bcbc31b7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-24 09:29:09.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/21/upstream-check/cinder-tempest-plugin-ontap-iscsi/f8da50e/ : RETRY_LIMIT in 7m 14s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/847131/21/upstream-check/cinder-tempest-plugin-ontap-nfs/dadd676/ : RETRY_LIMIT in 3m 12s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":21},{"id":"814866e8c218582cb500d5b5ed0a79930bfdaba6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-24 09:32:19.000000000","message":"Patch Set 21:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/785/.","accounts_in_message":[],"_revision_number":21},{"id":"1b2585a9d584d33b822adbe024083ff199790761","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-24 11:01:15.000000000","message":"Patch Set 21:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/847131/21/check/DellEMC_Unity_FC/fb83d72/DellEMC_Unity_FC/2290 : SUCCESS in 1h 35m 55s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/21/check/DellEMC_Unity_iSCSI/6e690f8/DellEMC_Unity_iSCSI/2297 : SUCCESS in 1h 51m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":21},{"id":"7571e6be6e5a40025aa01b4ea767dc24041c2959","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-24 11:10:09.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/31/847131/21/check/DellEMC_PowerStore_iSCSI/0b0fd4e/DellEMC_PowerStore_iSCSI/1539 : SUCCESS in 1h 25m 20s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/31/847131/21/check/DellEMC_PowerStore_FC/4a32388/DellEMC_PowerStore_FC/1735 : SUCCESS in 1h 08m 18s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/31/847131/21/check/DellEMC_PowerStore_NFS/e527e3a/DellEMC_PowerStore_NFS/1233 : FAILURE in 1h 20m 13s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/31/847131/21/check/DellEMC_PowerStore_NVMeTCP/442bde2/DellEMC_PowerStore_NVMeTCP/1092 : SUCCESS in 2h 00m 51s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/31/847131/21/check/DellEMC_PowerStore_NVMeFC/d8d63ff/DellEMC_PowerStore_NVMeFC/56 : SUCCESS in 1h 25m 11s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":21},{"id":"21751831afcac9eea6771f00d0bf127dd2c9ae09","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-24 11:27:02.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/31/847131/21/check/DellEMC_PowerFlex_3.6/a54508c/DellEMC_PowerFlex_3.6/1778 : SUCCESS in 1h 47m 13s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/31/847131/21/check/DellEMC_PowerFlex_4.5/70678d1/DellEMC_PowerFlex_4.5/657 : SUCCESS in 2h 17m 43s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/31/847131/21/check/DellEMC_PowerFlex_4.x_NVMeTCP/24ddfee/DellEMC_PowerFlex_4.x_NVMeTCP/615 : FAILURE in 14m 48s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":21},{"id":"513d8c3819b2e1b8e499f6040224e1290c02ecae","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2024-07-24 14:15:21.000000000","message":"Patch Set 21: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"7f12ab3a1cda764321a440e986a8343ea956dd0f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-07-25 01:09:47.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"c0c6b10fd4b91478ec9511ec27655f718eb94693","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 01:10:34.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"fda6019f45d885caf0016f39ec76982a15c74def","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 02:53:00.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70bea9b29bb244ea9938d59374e6eaed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e574535edbc47eb84b1e50e1076ea76 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4469c9563f5f48c0a544055a053c0a9c : SUCCESS in 10m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19220b492ad74ab9ac89c0e8ca526636 : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39b6469b6611433a9a9a189bc81ae695 : SUCCESS in 12m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/1f4877c2809f450584bd02ea9c05d6c8 : SUCCESS in 1h 07m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f0e9f414dee346ff9d63c3d72bb8be5b : SUCCESS in 1h 34m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08fab1c8071d4ef0b48e1cc4ba1988f8 : SUCCESS in 1h 01m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/591a301a744a46ef800dc1af231d01b7 : SUCCESS in 11m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3432380e0a2e48a7874487f55f3b01c1 : SUCCESS in 49m 24s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d20f4f412a6c48d1bc7ba045d9f867f9 : SUCCESS in 1h 39m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34f1aec36a914976998de78f7a41605a : SUCCESS in 35m 18s","accounts_in_message":[],"_revision_number":21},{"id":"3e023000afd53328e35573781bccc1db74d23532","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 02:53:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":21},{"id":"7fd8003857888c5f4128eb4705ea638b4993dc7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 02:54:55.000000000","message":"Patch Set 21:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0296244ec234fe2ad2728f572ee73e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18ddc7b4ed7a498fb019fe7f345f8b87 : SUCCESS in 1m 27s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60bfac5fd6d2407aa1f3ae9e2a09c017 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":21},{"id":"adb206f28196a2b8e1853c3ff6e7da38a3d4fc5b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-25 04:40:34.000000000","message":"Patch Set 21:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/21/20240725050010/NEC-V-ISCSI : SUCCESS in 4h 30m 20s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/847131/21/20240725050010/NEC-V-FC : FAILURE in 8h 40m 21s","accounts_in_message":[],"_revision_number":21},{"id":"5d5f07721afb7267336d84474e0f1061b0512554","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-31 18:48:20.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"b37bb6b92e5c6d46f056575de2ed89ee069fb6d5","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2024-08-14 11:19:11.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"4f31e2f2737b6777bd576f72216bf11b5a73096f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-09-25 14:28:46.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"4ab0500a1f1c9f8374260671253e9f2d1cad75eb","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2024-10-31 09:04:25.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"2249c72d3e6acafe7bb383cafa35b85588ba9031","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-11-01 04:00:08.000000000","message":"Patch Set 21:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/847131/21/fc : SUCCESS in 0h 56m 14s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/847131/21/iscsi : SUCCESS in 0h 51m 51s","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"f2427b461867e6c32a3e5573666ead748cd5e1b4","revisions":{"8fafd470b34b4b94d0e6e65f1a6049232f06b9f8":{"kind":"REWORK","_number":1,"created":"2022-06-22 07:58:43.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/31/847131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/1"}}},"commit":{"parents":[{"commit":"8f0964a893b98faa6f8fc751a67601c6b58366ed","subject":"Send the correct location URI to the Glance v2 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f0964a893b98faa6f8fc751a67601c6b58366ed"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:55:02.000000000","tz":180},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nTODO: document this capability a bit more.\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8fafd470b34b4b94d0e6e65f1a6049232f06b9f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8fafd470b34b4b94d0e6e65f1a6049232f06b9f8"}]},"branch":"refs/heads/master"},"bca6a2871ac3ba2a4afc6a05b23b2667e9833057":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-06-24 07:50:17.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/31/847131/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/2"}}},"commit":{"parents":[{"commit":"a363693058228d83854f84b52371b36298668689","subject":"Send the correct location URI to the Glance v2 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a363693058228d83854f84b52371b36298668689"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-24 07:50:01.000000000","tz":180},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nTODO: document this capability a bit more.\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bca6a2871ac3ba2a4afc6a05b23b2667e9833057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bca6a2871ac3ba2a4afc6a05b23b2667e9833057"}]},"branch":"refs/heads/master"},"b2d6647cbe6adb2eb8815205cfee9732a1f7e5cd":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-08-01 13:13:55.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/31/847131/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/3"}}},"commit":{"parents":[{"commit":"4af3e86c6eefc152f505f6961082644f5ee18e68","subject":"Send the correct location URI to the Glance v2 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4af3e86c6eefc152f505f6961082644f5ee18e68"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-08-01 13:13:42.000000000","tz":180},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nTODO: document this capability a bit more.\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b2d6647cbe6adb2eb8815205cfee9732a1f7e5cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b2d6647cbe6adb2eb8815205cfee9732a1f7e5cd"}]},"branch":"refs/heads/master"},"c80b13dc0f9bc6af959301d92e99295561dc2873":{"kind":"REWORK","_number":4,"created":"2022-08-02 08:04:22.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/31/847131/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/4"}}},"commit":{"parents":[{"commit":"4af3e86c6eefc152f505f6961082644f5ee18e68","subject":"Send the correct location URI to the Glance v2 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4af3e86c6eefc152f505f6961082644f5ee18e68"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-08-02 08:04:04.000000000","tz":180},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c80b13dc0f9bc6af959301d92e99295561dc2873"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c80b13dc0f9bc6af959301d92e99295561dc2873"}]},"branch":"refs/heads/master"},"b10b10f930edbc66e24c82012ede697f379e4241":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-01-16 14:56:42.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/31/847131/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/5"}}},"commit":{"parents":[{"commit":"08e80390f3d83f687552d268e6378fdfcddc0d34","subject":"Send the correct location URI to the Glance v2 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08e80390f3d83f687552d268e6378fdfcddc0d34"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-01-16 14:34:22.000000000","tz":120},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b10b10f930edbc66e24c82012ede697f379e4241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b10b10f930edbc66e24c82012ede697f379e4241"}]},"branch":"refs/heads/master"},"bbfffa38d7c87d4b83c42bcc92f50dc9fd028d73":{"kind":"REWORK","_number":6,"created":"2023-01-26 12:35:12.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/31/847131/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/6"}}},"commit":{"parents":[{"commit":"7d250602437978afc27725d6c93e11d55b72fab2","subject":"Merge \"Send the correct location URI to the Glance v2 API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d250602437978afc27725d6c93e11d55b72fab2"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-01-26 12:33:35.000000000","tz":120},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bbfffa38d7c87d4b83c42bcc92f50dc9fd028d73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bbfffa38d7c87d4b83c42bcc92f50dc9fd028d73"}]},"branch":"refs/heads/master"},"b09cc1fd5e5f9dbdbefb6966d4228e878a2fec12":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-02-08 13:33:35.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/31/847131/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/7"}}},"commit":{"parents":[{"commit":"eec61650ea01544c24151a656ef78bcccb47390f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eec61650ea01544c24151a656ef78bcccb47390f"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-02-08 13:33:11.000000000","tz":120},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b09cc1fd5e5f9dbdbefb6966d4228e878a2fec12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b09cc1fd5e5f9dbdbefb6966d4228e878a2fec12"}]},"branch":"refs/heads/master"},"c6e23b7eb83d0b0a852d46f0fe87669fe9c0ea81":{"kind":"REWORK","_number":8,"created":"2023-02-13 09:20:32.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/31/847131/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/8"}}},"commit":{"parents":[{"commit":"eec61650ea01544c24151a656ef78bcccb47390f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eec61650ea01544c24151a656ef78bcccb47390f"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-02-13 09:10:03.000000000","tz":120},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6e23b7eb83d0b0a852d46f0fe87669fe9c0ea81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6e23b7eb83d0b0a852d46f0fe87669fe9c0ea81"}]},"branch":"refs/heads/master"},"c47c8054fc1f6239f050a0865d8ac8f8da1b073d":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-03-01 12:49:08.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/31/847131/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/9"}}},"commit":{"parents":[{"commit":"9f7cb571bb95fec99f6113fc87785b3a271d5fc2","subject":"Merge \"Tests: Fix typo\u0027d unit test name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9f7cb571bb95fec99f6113fc87785b3a271d5fc2"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-03-01 12:49:01.000000000","tz":120},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c47c8054fc1f6239f050a0865d8ac8f8da1b073d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c47c8054fc1f6239f050a0865d8ac8f8da1b073d"}]},"branch":"refs/heads/master"},"eeb95e0adfda075e395a15c1825448a92824008d":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-05-08 19:32:02.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/31/847131/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/10"}}},"commit":{"parents":[{"commit":"40a33360ade30abb1656aa3762051a0d96c76186","subject":"Merge \"[Pure Storage] Add check for new error message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40a33360ade30abb1656aa3762051a0d96c76186"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-05-08 19:16:43.000000000","tz":180},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eeb95e0adfda075e395a15c1825448a92824008d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eeb95e0adfda075e395a15c1825448a92824008d"}]},"branch":"refs/heads/master"},"8ce9bb75fb32c921043b14b43884c8b1660cd9b8":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-12-06 06:26:30.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/31/847131/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/11"}}},"commit":{"parents":[{"commit":"842b39e2e8e73ea2dda84955d3ab99d9efbfa527","subject":"StorPool: create_cloned_volume() improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/842b39e2e8e73ea2dda84955d3ab99d9efbfa527"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-12-06 06:19:21.000000000","tz":120},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ce9bb75fb32c921043b14b43884c8b1660cd9b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ce9bb75fb32c921043b14b43884c8b1660cd9b8"}]},"branch":"refs/heads/master"},"c227eb238e464e31f3a08a1bdb2bab2bd844d50e":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2024-01-18 12:56:47.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/31/847131/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/12"}}},"commit":{"parents":[{"commit":"27a21230f5c4e78c2893c6425544558f8791a7c5","subject":"StorPool: create_cloned_volume() improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27a21230f5c4e78c2893c6425544558f8791a7c5"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2024-01-18 09:26:10.000000000","tz":120},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c227eb238e464e31f3a08a1bdb2bab2bd844d50e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c227eb238e464e31f3a08a1bdb2bab2bd844d50e"}]},"branch":"refs/heads/master"},"7431c9b36400cd86810e6c211e87019375fbd923":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-07-08 16:40:11.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/31/847131/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/13"}}},"commit":{"parents":[{"commit":"9b35fc5fa80d95d310de8716309aa378e0e3d7ec","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b35fc5fa80d95d310de8716309aa378e0e3d7ec"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-07-08 15:01:25.000000000","tz":0},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7431c9b36400cd86810e6c211e87019375fbd923"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7431c9b36400cd86810e6c211e87019375fbd923"}]},"branch":"refs/heads/master"},"3f25a59f4cc32c19132688692b4838ddf321f7bf":{"kind":"REWORK","_number":14,"created":"2024-07-09 05:17:44.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/31/847131/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/14"}}},"commit":{"parents":[{"commit":"4f29c54f964c625a2a9e0a92219f8c1b71198f86","subject":"Merge \"Tests: Fix misspelled scheduler unit test name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4f29c54f964c625a2a9e0a92219f8c1b71198f86"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-07-09 05:17:25.000000000","tz":330},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f25a59f4cc32c19132688692b4838ddf321f7bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f25a59f4cc32c19132688692b4838ddf321f7bf"}]},"branch":"refs/heads/master"},"1adf6ff1085f6246697159f4a38aca395a043d6d":{"kind":"REWORK","_number":15,"created":"2024-07-18 05:30:22.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/31/847131/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/15"}}},"commit":{"parents":[{"commit":"4f29c54f964c625a2a9e0a92219f8c1b71198f86","subject":"Merge \"Tests: Fix misspelled scheduler unit test name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4f29c54f964c625a2a9e0a92219f8c1b71198f86"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-07-18 05:23:43.000000000","tz":330},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1adf6ff1085f6246697159f4a38aca395a043d6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1adf6ff1085f6246697159f4a38aca395a043d6d"}]},"branch":"refs/heads/master"},"271750f168bc478b54bc4dc2d89805c97bb4db88":{"kind":"REWORK","_number":16,"created":"2024-07-20 15:56:08.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/31/847131/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/16"}}},"commit":{"parents":[{"commit":"4f29c54f964c625a2a9e0a92219f8c1b71198f86","subject":"Merge \"Tests: Fix misspelled scheduler unit test name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4f29c54f964c625a2a9e0a92219f8c1b71198f86"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-07-20 15:54:50.000000000","tz":330},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/271750f168bc478b54bc4dc2d89805c97bb4db88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/271750f168bc478b54bc4dc2d89805c97bb4db88"}]},"branch":"refs/heads/master"},"43d5fa823833c1aaa93bbc1b0b9aaf7eb1b3a395":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2024-07-23 04:46:03.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/31/847131/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/17"}}},"commit":{"parents":[{"commit":"834d933fa9fef6630d5df7ae278c6dad6dd82b80","subject":"Pass with mypy 1.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/834d933fa9fef6630d5df7ae278c6dad6dd82b80"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-07-23 04:18:38.000000000","tz":330},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43d5fa823833c1aaa93bbc1b0b9aaf7eb1b3a395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/43d5fa823833c1aaa93bbc1b0b9aaf7eb1b3a395"}]},"branch":"refs/heads/master"},"eb2338f3d9420a990e44311398d76231c91c4a5a":{"kind":"REWORK","_number":18,"created":"2024-07-23 07:25:23.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/31/847131/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/18"}}},"commit":{"parents":[{"commit":"834d933fa9fef6630d5df7ae278c6dad6dd82b80","subject":"Pass with mypy 1.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/834d933fa9fef6630d5df7ae278c6dad6dd82b80"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-07-23 07:25:07.000000000","tz":330},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb2338f3d9420a990e44311398d76231c91c4a5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb2338f3d9420a990e44311398d76231c91c4a5a"}]},"branch":"refs/heads/master"},"9b400d01660be47e73a2de7e002fdfcd4d2984b9":{"kind":"REWORK","_number":19,"created":"2024-07-23 11:25:56.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/31/847131/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/19"}}},"commit":{"parents":[{"commit":"834d933fa9fef6630d5df7ae278c6dad6dd82b80","subject":"Pass with mypy 1.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/834d933fa9fef6630d5df7ae278c6dad6dd82b80"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-07-23 11:21:58.000000000","tz":330},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b400d01660be47e73a2de7e002fdfcd4d2984b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b400d01660be47e73a2de7e002fdfcd4d2984b9"}]},"branch":"refs/heads/master"},"68edc19ad1234de6c2c464af0caebcb34bc7cb58":{"kind":"REWORK","_number":20,"created":"2024-07-24 06:42:54.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/31/847131/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/20"}}},"commit":{"parents":[{"commit":"834d933fa9fef6630d5df7ae278c6dad6dd82b80","subject":"Pass with mypy 1.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/834d933fa9fef6630d5df7ae278c6dad6dd82b80"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-07-24 06:42:05.000000000","tz":330},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different pool and\nrelax the checks when creating a volume from an image (both with and\nwithout the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68edc19ad1234de6c2c464af0caebcb34bc7cb58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68edc19ad1234de6c2c464af0caebcb34bc7cb58"}]},"branch":"refs/heads/master"},"f2427b461867e6c32a3e5573666ead748cd5e1b4":{"kind":"NO_CODE_CHANGE","_number":21,"created":"2024-07-24 06:44:29.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/31/847131/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/847131/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/847131/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/847131/21"}}},"commit":{"parents":[{"commit":"834d933fa9fef6630d5df7ae278c6dad6dd82b80","subject":"Pass with mypy 1.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/834d933fa9fef6630d5df7ae278c6dad6dd82b80"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:04:31.000000000","tz":180},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-07-24 06:43:18.000000000","tz":330},"subject":"Add the clone_across_pools driver capability","message":"Add the clone_across_pools driver capability\n\nLet drivers declare that they can clone a volume into a different\npool and relax the checks when creating a volume from an image\n(both with and without the image cache).\n\nChange-Id: Idbac4dae00f9fa03fb14547a204666721bf67d96\nImplements: blueprint clone-across-pools\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f2427b461867e6c32a3e5573666ead748cd5e1b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f2427b461867e6c32a3e5573666ead748cd5e1b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
