)]}'
{"id":"openstack%2Fcinder~847132","triplet_id":"openstack%2Fcinder~master~I5338c6c4f53a448e495f695cd64b36b722cd947d","project":"openstack/cinder","branch":"master","topic":"bp/clone-across-pools","attention_set":{},"removed_from_attention_set":{"33648":{"account":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"last_update":"2024-07-31 16:49:03.000000000","reason":"Change was submitted"},"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2024-07-31 16:49:03.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-07-29 21:30:01.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-07-25 12:58:06.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"35075":{"account":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"last_update":"2024-07-31 16:49:03.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2023-06-02 06:10:08.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2024-07-31 16:49:03.000000000","reason":"Change was submitted"},"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2024-07-31 16:49:03.000000000","reason":"Change was submitted"},"32425":{"account":{"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},"last_update":"2023-06-02 06:10:08.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-07-31 14:32:00.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"12988":{"account":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"last_update":"2024-07-31 16:49:03.000000000","reason":"Change was submitted"},"7198":{"account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"last_update":"2024-07-31 16:49:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5338c6c4f53a448e495f695cd64b36b722cd947d","subject":"StorPool: declare the clone_across_pools capability","status":"MERGED","created":"2022-06-22 07:58:43.000000000","updated":"2024-07-31 16:51:11.000000000","submitted":"2024-07-31 16:49:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847132-bp/clone-across-pools","meta_rev_id":"f3140c69957cbe7044f95731431c7a741906da7d","_number":847132,"virtual_id_number":847132,"owner":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-31 16:49:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"all":[{"value":0,"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2024-07-25 12:58:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":2,"date":"2024-07-29 21:30:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},{"value":2,"date":"2024-07-31 14:32:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},{"value":1,"date":"2024-07-31 14:32:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-22 08:12:22.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-06-22 08:33:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-22 10:42:47.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-06-22 11:07:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-22 11:07:17.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-06-22 11:53:29.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-22 12:16:59.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-06-22 12:28:40.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-06-22 13:12:56.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-22 13:27:50.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-06-22 13:54:52.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-22 16:54:15.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-06-22 17:13:34.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-06-22 20:40:07.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-22 23:20:56.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-06-23 02:21:41.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-06-23 06:49:20.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-06-23 07:57:22.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-06-23 08:02:36.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-23 08:09:23.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-06-23 09:27:26.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-06-27 14:42:17.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-15 14:49:11.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2022-07-16 05:27:47.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-07-20 03:42:11.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-01 18:26:24.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-01 21:38:06.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-02 06:11:34.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-08-02 17:49:53.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-08-03 07:00:30.000000000","updated_by":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"reviewer":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"state":"REVIEWER"},{"updated":"2023-01-17 10:38:43.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-19 06:50:17.000000000","updated_by":{"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},"reviewer":{"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},"state":"REVIEWER"},{"updated":"2023-01-26 12:52:43.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-01-26 16:05:03.000000000","updated_by":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"reviewer":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"state":"CC"},{"updated":"2023-02-28 10:29:54.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2023-03-08 22:23:27.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2023-05-08 22:31:36.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-05-19 14:23:08.000000000","updated_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"reviewer":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"state":"REVIEWER"},{"updated":"2023-05-19 15:15:12.000000000","updated_by":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"reviewer":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"state":"REVIEWER"},{"updated":"2023-06-02 06:10:08.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2023-06-02 06:17:55.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2024-07-25 07:33:23.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-07-25 10:47:58.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-07-25 10:50:52.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"CC"},{"updated":"2024-07-25 16:04:38.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-07-25 16:04:38.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2024-07-25 16:04:38.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"}],"messages":[{"id":"e6c898be0588d0d70225ed4f3c5ad1f565fdd3d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2022-06-22 07:58:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f0c5d71f8d443be8400ce9c711f08f9cae86214","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-22 08:12:22.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1655884733140 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"71e09e2f55f38d9374e7636756d8a2bb3f3b352e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 08:33:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/25305c4d732c40e5828958bad70b8b6c : SUCCESS in 22m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b68159b2b0d3455da4ba4ce4d8aa39d0 : FAILURE in 24m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be66ecbf3710daec0ca54d91c56597eedb6a8e49","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-22 10:42:47.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/847132/1/iscsi/ : SUCCESS in 1h 10m 26s\n- cinder-smb http://cloudbase-ci.com/cinder/847132/1/smb/ : FAILURE in 1h 08m 09s","accounts_in_message":[],"_revision_number":1},{"id":"af45be7abe225784df9afdd7b9d6bafba6b1bf9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 11:07:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f238bfd833ef461d8f1f693fd58e7ee6 : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f355d610a4a140878ca5858979adc09c : SUCCESS in 10m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62f178e3b42342c4a095029d805755f3 : SUCCESS in 11m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c5b9e95d59b498ba77dd593999b5239 : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20ce7ae4ac9244e58696a3bd26f83eb4 : SUCCESS in 11m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/fbc3ec75e066450a8d47e1daadd4e48f : SUCCESS in 1h 03m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/874da7d433a74a3d8def0abafe41f403 : SUCCESS in 1h 33m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8a3e5f2c42a47d6949162063e594a90 : SUCCESS in 49m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/afb63bd706d64a7ab9056cffed1c7893 : SUCCESS in 14m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/85e49ddcbac64c788d5b4210dad01c3d : SUCCESS in 5m 27s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/10ba55dff0d144ae8d124864db222af9 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4712b3c29b68414bbec203be49c691b4 : SUCCESS in 7m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/37e8790bdffe4710bbb259440405490e : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/65007693c94a40b1bcbc9e5841b16b98 : SUCCESS in 40m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9a07afd5d24846f18fd1389df5287d1b : FAILURE in 7m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/35371679c0984e8bb2b32da29eb4f119 : SUCCESS in 1h 23m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/985cdb6654cc4b21b826ca17fdfd612e : FAILURE in 1h 40m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e8049e7e8d2e4e07b49656b73721a5e9 : SUCCESS in 1h 32m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/735bbb22f0a5417aac8a7f3f3cc1207d : SUCCESS in 1h 10m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/770c56398ecf4a67a3bfddd769a19973 : SUCCESS in 1h 20m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b3a6f7a5bcd540089932fa20f4ed0fbc : SUCCESS in 38m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bb2352e3fd5a40e8b908ef2556ce4bbc : FAILURE in 1h 16m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8e36d2e04e444b6184e31d17f0a8518b : SUCCESS in 2h 49m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4e7860a06c34714b90556452a9a8bc3 : SUCCESS in 56m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5a0ef022fc84245e2a308347a15589943158e7e8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-06-22 11:07:17.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"a5620a91eec9230317349fb079853fcef8ed9321","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-22 11:53:29.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/9a147625e8154f5b831b68283f9e0dc7 : POST_FAILURE in 2h 06m 45s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/687ebdbc007c4ec292da320754c84e04 : POST_FAILURE in 2h 44m 32s","accounts_in_message":[],"_revision_number":1},{"id":"2c91f2aaf3870895847aff3d319fb239cd34b80b","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-06-22 12:16:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-32-847132-1/1655896027 : SUCCESS in 17m 05s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":1},{"id":"bc3673391eba15cb0cc77795baa8d2a12b068c1d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-22 12:28:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k-focal https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/847132/1/check/iscsi-driver-alletra-6k-focal/46b7ea1/46b7ea1.html : FAILURE in 56m 57s","accounts_in_message":[],"_revision_number":1},{"id":"d96486ec5735d536ace68c37fac5e6fcdbb2880b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-22 13:12:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e324178cd63146acaffcb2a16ad06ac1 : SUCCESS in 38m 43s","accounts_in_message":[],"_revision_number":1},{"id":"bad050f05ab3842ae602aef24102bd13c475fcab","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-22 13:27:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/1/upstream-check/cinder-tempest-plugin-ontap-fcp/f49465d/ : SUCCESS in 1h 01m 59s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/b1bd8d9/ : FAILURE in 45m 15s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/1/upstream-check/cinder-tempest-plugin-ontap-nfs/4ccbbe1/ : SUCCESS in 40m 18s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"4d729efe42c851b2fdaa62f2ba26c2256730e8e8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-22 13:54:52.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-847132-1/ns5_nfs : SUCCESS in 3h 35m 21s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-847132-1/ns5_iscsi : SUCCESS in 3h 40m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"3c6d178b26c14310a8859120e9ee683d4e61ebce","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-22 16:16:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/32/847132/1/check/iscsi-driver-nimble/87f1f31/87f1f31.html : FAILURE in 47m 56s","accounts_in_message":[],"_revision_number":1},{"id":"61ea14be9de68eaf7b56e3a3a2418a9ae3618ac9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-22 16:54:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/32/847132/1/check/dsvm-tempest-cisco-zonemanager-job/e483e00/ : SUCCESS in 1h 10m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7f928e7985b014e89852cf2fabb01faf24a72ae1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-22 17:13:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/32/847132/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/50c659b/ : TIMED_OUT in 3h 02m 43s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/32/847132/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0e0a03e/ : TIMED_OUT in 3h 03m 45s","accounts_in_message":[],"_revision_number":1},{"id":"616eec3402b7da19bd030084709cb6993b90e979","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-06-22 20:40:07.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/32/847132/1/check/DellEMC_PowerFlex/4516578/DellEMC_PowerFlex/888 : SUCCESS in 1h 34m 09s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"33f80859fb09f67d5f85c91352cd1c47f56ecbd8","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-06-22 23:20:56.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/32/847132/1/check/DellEMC_SC_FC/d78cf54/DellEMC_SC_FC/734 : SUCCESS in 58m 12s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/1/check/DellEMC_SC_iSCSI/5cea10d/DellEMC_SC_iSCSI/756 : SUCCESS in 1h 52m 23s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"3202188146dc0cfc988759807f3acff3f929ddf2","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-06-23 02:21:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/847132-1/HPMSA-iSCSI/ : FAILURE in 53:56.49\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/847132-1/HPMSA-FC/ : FAILURE in 50:27.12","accounts_in_message":[],"_revision_number":1},{"id":"b7c69fc8235a70af7c336b8bba5e2f0094d2f7c2","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-06-23 06:49:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/847132-1/PVME-iSCSI/ : FAILURE in 1:10:54\n- PVME-FC https://os-logs.seagate.com/dell-ci/847132-1/PVME-FC/ : FAILURE in 1:04:05","accounts_in_message":[],"_revision_number":1},{"id":"1c22a4badfbc7d7b0b333ef59a63b9884e5e0e33","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-23 07:57:22.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/847132/1/check/DellEMC_Unity_FC/81a049b/DellEMC_Unity_FC/1001 : SUCCESS in 1h 10m 11s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/1/check/DellEMC_Unity_iSCSI/ce9e07c/DellEMC_Unity_iSCSI/941 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"91cb32818792b9aa798fcaf4ac3e4f7e7760e392","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-06-23 08:02:36.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"4dd3aaea48366bba72a9c735bfe356c9a981c966","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-06-23 08:09:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/847132-1/Lenovo-iSCSI/ : FAILURE in 1:05:34\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/847132-1/Lenovo-FC/ : FAILURE in 58:52.32","accounts_in_message":[],"_revision_number":1},{"id":"6e2ee2c5748e54612ed465243b7b8cff5be84731","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-06-23 09:27:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/847132-1/Seagate-iSCSI/ : FAILURE in 1:11:22\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/847132-1/Seagate-FC/ : FAILURE in 1:08:02","accounts_in_message":[],"_revision_number":1},{"id":"41f0ff5a22b30bb8f99fd64c3a7d92ec665f11f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2022-06-24 07:50:17.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"50f6668373fe4af2c5f29841261ee68c49151746","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 08:12:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8731da937b843d3a65112cfc721155c : SUCCESS in 17m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3bd1529adac2456d9e57be041c01f64d : SUCCESS in 19m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0b17f987f9e31d0483acac28c5eebc2fe74e4b65","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-24 08:21:45.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"22551a5e86acb0b228ebcdd0d11ee539d537f715","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-24 10:21:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k-focal https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/847132/2/check/iscsi-driver-alletra-6k-focal/382e8f7/382e8f7.html : FAILURE in 54m 29s","accounts_in_message":[],"_revision_number":2},{"id":"7120158d07052aeac1bb8378fb259a279f9f131b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-06-24 10:33:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/847132/2/iscsi/ : SUCCESS in 1h 03m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/847132/2/smb/ : SUCCESS in 58m 07s","accounts_in_message":[],"_revision_number":2},{"id":"08ca51170711047768c54eb0243d2383645003a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 11:12:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4567a557d4244fbb952e1cf8ffa503bf : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8f04553c46049dd81b207250f19710f : SUCCESS in 10m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/309e8642d71d4e32aaa0052d1c834471 : SUCCESS in 11m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a4bd18b060e4781810fb5fff7f95754 : SUCCESS in 11m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aa21975ef3a4073bb751b11f02c918a : SUCCESS in 13m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/3137956f9ff94946bd8f79b91dd13fd0 : SUCCESS in 52m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e5833ace9dca4eaeb53e3c689b00ee87 : SUCCESS in 1h 14m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23e7af49f095410ba52e1a9315089f56 : SUCCESS in 56m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/aa5468d8677f4ee4b5dd4b2cbdc9b100 : SUCCESS in 18m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e170d239fcd442ea9223970b0086e70b : SUCCESS in 6m 11s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/50cac8b132a942f09223a3fc33f6e23e : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9f9d782efa944d839db609e5c3ae1a10 : SUCCESS in 6m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/94a9cc5a84fd4988a17012dffeb34ba7 : SUCCESS in 7m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/28247f9bfecc4e9095c54711ebdbe929 : SUCCESS in 36m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5950ee792c444c0cb59f7837a84a9c3b : FAILURE in 7m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/47be0ea143fa4a39a50e694dc193c5e6 : SUCCESS in 1h 39m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ad42e6941c364b66836439dce77a0034 : FAILURE in 1h 23m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c33b80f4b20e41b38bed65b5c622fe54 : SUCCESS in 1h 39m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fc33897fc0e347459b38e51b2e10dcda : SUCCESS in 1h 16m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8c34b673cc43454e8dc6ec3d3f0d5d71 : SUCCESS in 1h 37m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/37c93bd3e9a64577885aa2d3a5763eda : SUCCESS in 39m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/788c0d7857dc4253ba98fded5194cdc3 : FAILURE in 1h 15m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fe6756cf86864736a34498a2127c81f9 : SUCCESS in 2h 12m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25a49ee74ec64628bb61718516343c75 : SUCCESS in 1h 01m 25s","accounts_in_message":[],"_revision_number":2},{"id":"068f6a2def282974fd25e754e969aebfdef06a67","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-06-24 11:13:09.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"7f1817e07521d0940a3c90a3ee6d2fb6cc328725","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-06-24 11:47:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-32-847132-2/1656069120 : SUCCESS in 16m 54s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"bf8d92f5fffa45d74600652a7ac2d759df58ccdc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-24 12:50:58.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/2/upstream-check/cinder-tempest-plugin-ontap-fcp/6afcdbc/ : SUCCESS in 1h 04m 36s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/fcca771/ : RETRY_LIMIT in 7m 03s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/2/upstream-check/cinder-tempest-plugin-ontap-nfs/333439c/ : SUCCESS in 41m 35s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"7678b716c3b83b9ce7404f2ebba855b3198d5c95","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-24 13:24:38.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-847132-2/ns5_nfs : SUCCESS in 3h 23m 41s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-847132-2/ns5_iscsi : SUCCESS in 3h 40m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"f17df3f615e7d24df3e32df36937742b8e123c0a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-24 13:51:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/32/847132/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d0c6add/ : SUCCESS in 1h 25m 51s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/32/847132/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7882508/ : FAILURE in 1h 20m 48s","accounts_in_message":[],"_revision_number":2},{"id":"cb38475817b67322feb2278fb38043026961e624","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-06-24 14:46:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/dcfb677df4c94c65b092c9779060ca24 : FAILURE in 1h 07m 23s","accounts_in_message":[],"_revision_number":2},{"id":"f9d7649ae9cf2b4cd2d82c6586569592d28e98ba","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-06-24 21:34:52.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/32/847132/2/check/DellEMC_PowerFlex/db7e788/DellEMC_PowerFlex/909 : ABORTED in 2h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"191784e3b8acc23ee14dd3fd188a6a9eb12e5389","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-06-24 22:31:59.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/32/847132/2/check/DellEMC_SC_FC/5e3c136/DellEMC_SC_FC/755 : SUCCESS in 47m 20s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/2/check/DellEMC_SC_iSCSI/fb3ac4a/DellEMC_SC_iSCSI/777 : SUCCESS in 1h 47m 27s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"37e3037f5f6dbe8702116a2058d1c97d9e455aca","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-06-24 22:52:56.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/32/847132/2/check/iscsi-driver-nimble/e310b64/e310b64.html : FAILURE in 50m 43s","accounts_in_message":[],"_revision_number":2},{"id":"0653cb759f34c2c74072385749bbc7d0a4f6cf76","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-06-25 03:06:44.000000000","message":"Patch Set 2:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/847132-2/HPMSA-iSCSI/ : FAILURE in 54:44.50\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/847132-2/HPMSA-FC/ : FAILURE in 52:13.55","accounts_in_message":[],"_revision_number":2},{"id":"37b36afaca2da2e3ce80d7ec5a3f4983d437109c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-06-25 11:35:50.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 1h 14m 41s","accounts_in_message":[],"_revision_number":2},{"id":"5798a56a826a42ff58d99729446aedbc0109beee","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-06-25 16:25:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/847132-2/Lenovo-iSCSI/ : FAILURE in 1:07:21\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/847132-2/Lenovo-FC/ : FAILURE in 1:04:48","accounts_in_message":[],"_revision_number":2},{"id":"4a298ec40f3a69c3a1af7c92b98fab637bfd9a63","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-25 17:28:50.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/847132/2/check/DellEMC_Unity_FC/8c1cba0/DellEMC_Unity_FC/1032 : SUCCESS in 1h 08m 59s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/2/check/DellEMC_Unity_iSCSI/87fca61/DellEMC_Unity_iSCSI/968 : SUCCESS in 2h 03m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"f3af650519d1c65ad6343f0c5b41963722484353","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-06-25 22:12:40.000000000","message":"Patch Set 2:\n\nBuild failed.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/847132-2/Seagate-iSCSI/ : FAILURE in 1:07:19\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/847132-2/Seagate-FC/ : FAILURE in 1:08:39","accounts_in_message":[],"_revision_number":2},{"id":"d1ca20aac4a6befa0795bd7f0d567e8e6c43f91c","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-06-27 14:42:17.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/32/847132/2/check/DellEMC_PowerMAX_FC/0f0018d/DellEMC_PowerMAX_FC/628 : SUCCESS in 3h 20m 20s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/2/check/DellEMC_PowerMAX_iSCSI/4c1208c/DellEMC_PowerMAX_iSCSI/615 : ABORTED in 1m 27s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"8385a56e89e94240627c2866156df2632dfa5b06","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-07-15 14:49:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8749f973fa7954dffab6a7f8f40295e0e9676a9f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-16 05:27:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/2/20220716120624/NEC-V-FC : SUCCESS in 2h 11m 04s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/2/20220716120624/NEC-V-ISCSI : SUCCESS in 2h 21m 20s","accounts_in_message":[],"_revision_number":2},{"id":"724c6670584164385941a8cfd2369ce649235845","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-07-20 03:42:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/2/20220720091143/NEC-ISCSI : SUCCESS in 3h 30m 24s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/2/20220720091143/NEC-FC : SUCCESS in 3h 30m 24s","accounts_in_message":[],"_revision_number":2},{"id":"3aa92e6da87cdde9047f633d173e1ed9dc8f7799","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2022-08-01 13:13:55.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"0c40f47b3d1dac6f5e540c9305ae8b33e2d8893e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-01 13:28:40.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"e6b6f46e3cd8775ebd6855484663ad2c47ed50ad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-01 13:53:47.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1659359645862 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"e88e222f4f5679d7323d2ec65234ce8370332d35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 14:20:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b303b4297d1a4c569f913527e7ad29a0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3536049fe0644539899c063de2eeb461 : SUCCESS in 20m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1c7265595e6a4897b0f1245e5b28d108 : SUCCESS in 23m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"781209d7d2105c991c262a040aa3df07d75030d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 15:30:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c27ae72d438c4113ad84db60ff4f7428\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/041101e581bf4d4a98e9f9bb89736122 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a818b45a421a43f384caa6ca54932d5a : SUCCESS in 11m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc658b65a7fc4738a92291e9f5adb793 : SUCCESS in 11m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0812386e651410daee3db9ae38f64eb : SUCCESS in 11m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d73bec491a9f4763ab7bd080b340882f : FAILURE in 7m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/a05e71eed73843be8eddc332da3b29ec : SUCCESS in 1h 11m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4c51d295ed3445ddbde367c0bf082b4a : SUCCESS in 1h 22m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5fa6906025f4c05bcd689682befd7ee : SUCCESS in 50m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/eff925771eca403294d9a77d58745e8b : SUCCESS in 15m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5179d4a962834d9ca5e00169c100daab : SUCCESS in 9m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f8895ab937df453d9c79ec5745c7dc44 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/560d9ee134d544ec89cf31e25c5f1f3a : SUCCESS in 6m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/828a7b0706f2404a9d4a254e0ab7b959 : SUCCESS in 7m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e94bedcde0ea4a2aa49e7c03d2256901 : SUCCESS in 55m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ab56756e757d4e0f888de72ac3039746 : FAILURE in 7m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1a374efbc61a46a7a0716fa6bb05e159 : SUCCESS in 1h 20m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9292688051314743aa944fa2519abc77 : FAILURE in 1h 31m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/36f23e3cc95c4134b09a8b13487144f9 : SUCCESS in 1h 49m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e03b13f5fbae490cacb80db8b431ea1f : SUCCESS in 1h 02m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7e6030ecdb884326957a006e4db1115e : SUCCESS in 1h 39m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5f7084de32ad49b28a5b0166df4f55c8 : SUCCESS in 46m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/57d187ce2d474c78a7a49c0e7d56796f : SUCCESS in 1h 43m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eddfd048d2fc423b961784bbcc509767 : SUCCESS in 2h 13m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/657eeb92fede4101a5fcda226bb3b0bb : SUCCESS in 55m 53s","accounts_in_message":[],"_revision_number":3},{"id":"b1790b30d0ddd7a9c60204fc052e3f011e4120d7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-01 16:09:16.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/847132/3/iscsi/ : SUCCESS in 1h 00m 30s\n- cinder-smb http://cloudbase-ci.com/cinder/847132/3/smb/ : SUCCESS in 1h 34m 04s","accounts_in_message":[],"_revision_number":3},{"id":"fd2d30a688e5a5e9d60897436e90246c16c73dbe","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-01 17:42:43.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/eac848005c7a4abca2c7cc475f7d0b8d : POST_FAILURE in 1h 42m 39s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/e69269d1db5044de91cbe8df8706262b : TIMED_OUT in 3h 07m 18s","accounts_in_message":[],"_revision_number":3},{"id":"b09bd113efb2363544101f6401a97fe826dfb44d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-01 18:20:37.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/847132/3/check/iscsi-driver-alletra-6k/c32133d/c32133d.html : FAILURE in 26m 18s","accounts_in_message":[],"_revision_number":3},{"id":"43e55c7f1dea5b5414eaf471e0bdb3d254e5d5c7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-01 18:26:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847132 : FAILURE in 1h 01m 18s","accounts_in_message":[],"_revision_number":3},{"id":"92a842e64a7ab6a1266b302e1799752e8ec0d221","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-01 19:12:06.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-847132-3/ns5_nfs : SUCCESS in 3h 26m 10s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-847132-3/ns5_iscsi : SUCCESS in 3h 27m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"812b8bda92ab801978b6d5042249e63c7aca4c9e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-08-01 19:18:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9b8043ab21c540b49d87cb8d4b7a3042 : SUCCESS in 46m 09s","accounts_in_message":[],"_revision_number":3},{"id":"1cb556fb061d3709f003a44e47c3520cbbfe5cf0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-01 19:22:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/32/847132/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/3e66548/ : SUCCESS in 1h 43m 22s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/32/847132/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/bee0389/ : SUCCESS in 1h 29m 23s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/32/847132/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/67ddd2c/ : SUCCESS in 1h 23m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f683b35bbbd31d38375df7f00346c5cbf156f093","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-01 21:38:06.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/32/847132/3/check/fujitsu-eternus-dx-iscsi/23a6e46 : SUCCESS in 1h 43m 57s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/32/847132/3/check/fujitsu-eternus-dx-fc/7d26db1 : FAILURE in 1h 54m 11s","accounts_in_message":[],"_revision_number":3},{"id":"8f2000cd7cfbed1b3d138a920d132634f5b3c531","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-02 03:46:10.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/847132/3/check/DellEMC_Unity_FC/114ed1d/DellEMC_Unity_FC/1324 : FAILURE in 20m 23s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/3/check/DellEMC_Unity_iSCSI/6ded9d4/DellEMC_Unity_iSCSI/1204 : SUCCESS in 2h 20m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"7c485813791154754049f3559a6d23ab40bc239f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-02 06:11:34.000000000","message":"Patch Set 3:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/347","accounts_in_message":[],"_revision_number":3},{"id":"e07feedaa7dc88afae42983ca8a2e48ebc5ee540","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2022-08-02 08:04:22.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"38aa34dddf26a8382ed9e9346dacfb4b09a3db6d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-02 08:18:10.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"4c1630cc8a83f70866567ceb0b5aac612e18076e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 08:28:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4dc9a1dbc0e491e9e9c8fb8e3580040\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/80cfd782753143bfb33a961885f04dd4 : SUCCESS in 20m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/986bf52b3e9344ca81853f35dfa9642d : SUCCESS in 21m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"514cf3269af3ed4827bbbf9d55809b1bf66afc03","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-02 09:29:01.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1659427473692 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"7ba71a30ad5b39f3bebfb5fef1a55fe644886c0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 10:14:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/615402b9cc1b430f9a3d98dcba879271\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5779a7ed9d614fbcbb9e531c47521817 : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50ea8efc850448418a209d3c1f7b049f : SUCCESS in 11m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef689fcedc034b508d0340813567fc21 : SUCCESS in 11m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7061bea51e24424a9e5c4cc0ffc8e53 : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28a191f094b84f3989ff3c8ee3d9b7e6 : SUCCESS in 12m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/6c18572eda684ebb8a0188e8fb9b79d8 : SUCCESS in 53m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/25434850ffea464f8b5b8e7913ac7427 : SUCCESS in 1h 26m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c71e8c7ce2f47ceb81b39b4564c6f9e : SUCCESS in 48m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d4bc58f6fedd4e889a6e77f64380bfaa : SUCCESS in 16m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b03f2e95c8f1490e8f5afb7a2094b86d : SUCCESS in 5m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0ae2f70f10624d61978f9ce9292616a2 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/746b77a2ef5d4e87af1fec1bca022621 : SUCCESS in 6m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b6e8270a918441bbb3befa79a678c723 : SUCCESS in 5m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/065c410e1b804fc5908d4be52ca45579 : SUCCESS in 36m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/78dac299778e4fee8976407d07b70c6f : FAILURE in 6m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/96de37e5e224497c93e2366356755711 : SUCCESS in 1h 05m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2bacc1e37dc44950a0ea05a7d14e4b35 : FAILURE in 1h 35m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5530703d4dd84774b8680e41ae239625 : SUCCESS in 1h 29m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e6e22aeb6c214ab5aaf2641596a232d4 : SUCCESS in 1h 09m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7302d6efb1e14cf298bb5811688d9413 : SUCCESS in 1h 16m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9d90ba9698f145b4b91ec1e90f41864e : SUCCESS in 46m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b9f14ffad04c4c698a99b108100fa269 : SUCCESS in 1h 21m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b33195152cb24228af0ea8b048e62205 : SUCCESS in 2h 07m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17c1c8aa305e431abdf4d980d56b0b8f : SUCCESS in 45m 16s","accounts_in_message":[],"_revision_number":4},{"id":"488f2461c76c60ddb9f2b5e6dddd9b980863c5e3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-02 10:35:18.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/847132/4/iscsi/ : SUCCESS in 1h 10m 57s\n- cinder-smb http://cloudbase-ci.com/cinder/847132/4/smb/ : SUCCESS in 1h 20m 46s","accounts_in_message":[],"_revision_number":4},{"id":"49645c46fd387d79a5af1cfb6958b190fe623ab8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-02 10:36:20.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/847132/4/check/iscsi-driver-alletra-6k/06ec5f3/06ec5f3.html : FAILURE in 52m 16s","accounts_in_message":[],"_revision_number":4},{"id":"80eb1f7610d8f3edcf8a6000af5600fed3317a57","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-02 10:41:15.000000000","message":"Patch Set 4:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/02_08_2022_10_33_22\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"9282d9005a3d7d7325c9c2a0841ca3d41a4b7723","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-02 10:59:11.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/4/upstream-check/cinder-tempest-plugin-ontap-fcp/d25b615/ : RETRY_LIMIT in 9m 44s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/c530a75/ : RETRY_LIMIT in 9m 53s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/4/upstream-check/cinder-tempest-plugin-ontap-nfs/ecb7b59/ : SUCCESS in 42m 30s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/4/upstream-check/cinder-tempest-plugin-solidfire-iscsi/cd749a7/ : RETRY_LIMIT in 9m 26s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"e634f1abd30f294a78937e24fd0a2ff5258e18f7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-02 12:16:49.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847132 : FAILURE in 1h 03m 36s","accounts_in_message":[],"_revision_number":4},{"id":"848eaf888586f22d9beab5cca586d30aef227a07","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-02 12:17:23.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/c4cf7e31f4a74d48acc31d5f282bfc88 : TIMED_OUT in 3h 11m 23s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/607e1bdf0e22443fb704fa5f045739f8 : TIMED_OUT in 3h 11m 30s","accounts_in_message":[],"_revision_number":4},{"id":"77d3694899a8dbda460c279ac578e4a8e46eadf6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-08-02 13:08:02.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e6ce313baae54314ababa8225076a755 : SUCCESS in 40m 57s","accounts_in_message":[],"_revision_number":4},{"id":"08e5a55b6decfa7e1fa12d5159dd7fa9485b5ae3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-02 13:31:25.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-847132-4/ns5_nfs : SUCCESS in 3h 31m 13s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-847132-4/ns5_iscsi : SUCCESS in 3h 33m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"bd1f5173ae7f160f6b88fc602f3550d03fd69c6b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-02 13:58:03.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/32/847132/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f6f2dae/ : SUCCESS in 1h 41m 38s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/32/847132/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/8158edf/ : SUCCESS in 1h 37m 39s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/32/847132/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/24ab5a0/ : SUCCESS in 1h 38m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5cfa934fff6a0386b5655a42ebfb293b065f693a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-02 14:29:37.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/32/847132/4/check/DellEMC_PowerFlex/b93be23/DellEMC_PowerFlex/1121 : FAILURE in 1h 04m 28s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"96c132d9c2a1b2ca94fd0c81e9f60c340c06a3d7","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-02 17:49:53.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/32/847132/4/check/DellEMC_PowerStore_FC/7450cd5/DellEMC_PowerStore_FC/660 : FAILURE in 17m 12s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/32/847132/4/check/DellEMC_PowerStore_NFS/8997568/DellEMC_PowerStore_NFS/204 : FAILURE in 19m 26s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/32/847132/4/check/DellEMC_PowerStore_NVMeTCP/87e78f1/DellEMC_PowerStore_NVMeTCP/76 : SUCCESS in 1h 56m 09s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"6e330d90d45ed5a5f31d756ecccfb0cc1c03c6fb","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-08-02 18:40:40.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/32/847132/4/check/DellEMC_SC_FC/a5058d9/DellEMC_SC_FC/850 : SUCCESS in 48m 12s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/4/check/DellEMC_SC_iSCSI/67915e3/DellEMC_SC_iSCSI/868 : SUCCESS in 2h 06m 49s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"f9ba5831eb3356dce414f43d5493273bd83adbc3","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-02 22:22:06.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/32/847132/4/check/fujitsu-eternus-dx-iscsi/c981378 : FAILURE in 3h 49m 13s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/32/847132/4/check/fujitsu-eternus-dx-fc/135ee17 : FAILURE in 3h 49m 21s","accounts_in_message":[],"_revision_number":4},{"id":"b8800063729686cd57a5f72935df5494afe3099c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-03 03:51:36.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/847132/4/check/DellEMC_Unity_FC/d32c52d/DellEMC_Unity_FC/1337 : FAILURE in 17m 59s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/4/check/DellEMC_Unity_iSCSI/3b54d8b/DellEMC_Unity_iSCSI/1216 : SUCCESS in 2h 12m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"3a2e8e0fefb842a0682b1de8afa7a762c923887f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-03 06:33:28.000000000","message":"Patch Set 4:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/356","accounts_in_message":[],"_revision_number":4},{"id":"a8b616d85c2fe3a089c6807256b8689b016b99eb","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-08-03 06:42:24.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/4/20220803132030/NEC-V-FC : SUCCESS in 2h 13m 18s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/4/20220803132030/NEC-V-ISCSI : SUCCESS in 2h 21m 51s","accounts_in_message":[],"_revision_number":4},{"id":"923ee2aded43280893a9eefe92f04a53149319de","author":{"_account_id":35159,"name":"李丹","email":"li.dan@99cloud.net","username":"Danee"},"date":"2022-08-03 07:00:30.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6d183db2c3d8b6726c120808e00e2c88c1a66185","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-08 17:46:52.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/4/20220808200754/NEC-ISCSI : SUCCESS in 3h 22m 38s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/4/20220808200754/NEC-FC : SUCCESS in 6h 38m 55s","accounts_in_message":[],"_revision_number":4},{"id":"cf49c9ff4f7834dab63e36822e41e32137cf9e03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-01-16 14:56:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9f758c3cb6ddc57937f22b70298b5ae56afc09b1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-16 15:01:02.000000000","message":"Patch Set 5:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a9e8fbfe95f2403f8e9e26ebfcdd695799f1c874","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-16 15:31:25.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"74baa09c4b87ecc1f2d48f196f6166ba70ae08be","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-16 15:47:07.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/32/847132/5/check/iscsi-driver-nimble/b58058b/b58058b.html : FAILURE in 14m 45s","accounts_in_message":[],"_revision_number":5},{"id":"a4ce7167d18c036a0d70afee5e691f4ecb1926ff","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-16 16:53:16.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/847132/5/check/iscsi-driver-alletra-6k/4cac157/4cac157.html : FAILURE in 25m 58s","accounts_in_message":[],"_revision_number":5},{"id":"dbcba0a7f11b16ca9209995b598d885b8244cf61","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-16 16:58:18.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/847132 : FAILURE in 23m 26s","accounts_in_message":[],"_revision_number":5},{"id":"149e779d3bbe3a2ad102a1dc4314e5f98b658d2e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-16 17:08:25.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847132 : FAILURE in 25m 57s","accounts_in_message":[],"_revision_number":5},{"id":"6afdc69315a398b726d7dd16bea89bfc084f7144","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:08:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/956dbdb73f584e61bea0e344cd58b004\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8894c063df4c4768a4200281c1fad6f6 : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aafa0051a29445378df8e9531cc9ccc6 : SUCCESS in 10m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0491e19c85fa489c8bd66e7dc9154001 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75a2edf1ed564e6ba638be7cdcef0674 : SUCCESS in 11m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/7989aea3dfd24b3c8408a3c59aa7c39a : SUCCESS in 1h 02m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bbe64080cec14ec48807bd1e5d4b2166 : SUCCESS in 46m 25s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b72586cad994cd0a4e006389a47f149 : SUCCESS in 1h 33m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34ddad8b9ef048a68260735f0eb2b77a : SUCCESS in 44m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c67f3296f37440ef8ad3bf119cc387a4 : SUCCESS in 15m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2f06781a323b45b48b1ccec096b7f1a4 : SUCCESS in 4m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f72bf916f8d84658850d674599c4506d : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2e586dedd9d546c98b836007bb91eb2c : SUCCESS in 5m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6fcf8eb3b32d48b291ebb95ff2220738 : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c664e92af1974cf68eac15ba305b4ec5 : SUCCESS in 47m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8e18c903d627487fb77029fc8878884a : FAILURE in 4m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8b8218ea6610447c82d3d7c1d2c3362f : SUCCESS in 1h 18m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9aad2702ad334a4389f9a0295f93de88 : FAILURE in 1h 14m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/10de06f38e744540a708fa607143211f : SUCCESS in 1h 36m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d451ee0047a743e7bc68a0f21dad8c15 : SUCCESS in 1h 08m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c5cd0d8678ea41eb83ce166c12002dd6 : SUCCESS in 1h 21m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b5d5f9026236400fa7f76071d7e881b3 : SUCCESS in 26m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2339604a3ff64a0ba3e230055ce23d13 : SUCCESS in 1h 34m 25s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0aad93ea51cc4be482437e32f43592e8 : SUCCESS in 2h 04m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cc83af514134dffa71eceb35f1745e9 : SUCCESS in 42m 01s","accounts_in_message":[],"_revision_number":5},{"id":"39cc522c4629fb91f26dd7396996a0731b97c3b3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-16 17:33:18.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-847132-5/ns5_nfs : FAILURE in 1h 51m 2s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-847132-5/ns5_iscsi : FAILURE in 37m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"3146742ad4a88726de829245d68c29a18b0c6262","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:35:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9823079033cc44699ba461a2212783fb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d56e6f248e34775808adba6ee043264 : SUCCESS in 26m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5189e381476447c7a34f18242c7fef51 : SUCCESS in 27m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e2b487f1cfa986088ac364b25b240c66af7d3cba","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-16 18:04:31.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/fa80c6598570440ca7c29d40fd964913 : POST_FAILURE in 1h 31m 45s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/d294bad89b9d4e0b8b7242c328e95011 : POST_FAILURE in 1h 20m 27s","accounts_in_message":[],"_revision_number":5},{"id":"5d11c19ddfc9413ebe2ed1d8826ed9d5c571722b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-16 18:08:06.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/5/upstream-check/cinder-tempest-plugin-ontap-fcp/27b7751/ : SUCCESS in 56m 30s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/1724a2d/ : SUCCESS in 43m 52s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/5/upstream-check/cinder-tempest-plugin-ontap-nfs/03984d4/ : SUCCESS in 44m 39s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"805182d9d213f206712ae07e739de9c976457677","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-16 20:12:30.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/32/847132/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/cef02c8/ : SUCCESS in 2h 12m 09s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/32/847132/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/54e66d1/ : FAILURE in 2h 48m 56s\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/32/847132/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/a7b2090/ : FAILURE in 2h 36m 51s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/32/847132/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/6e7a7ca/ : FAILURE in 2h 21m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b26abde76a9b6326d224026b4a0655bc97bcdea6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-01-16 20:28:54.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/32/847132/5/check/DellEMC_PowerFlex/2161d0a/DellEMC_PowerFlex/1525 : SUCCESS in 1h 21m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"1bf416c92e83e09ac2bdc204a9c989b9af63b92c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-16 23:57:49.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/847132/5/check/DellEMC_Unity_FC/cf469fd/DellEMC_Unity_FC/2157 : SUCCESS in 1h 57m 52s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/5/check/DellEMC_Unity_iSCSI/4290de6/DellEMC_Unity_iSCSI/2026 : ABORTED in 3h 00m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"22afa1afba6f93f5c7eba3f448486956fb8fae0d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-17 10:38:43.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/32/847132/5/ : FAILURE in 54m 54s","accounts_in_message":[],"_revision_number":5},{"id":"02d8fcc12a9346c53fa2987e172bebb4d881f2b8","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-01-18 06:02:40.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/5/20230118083943/NEC-V-ISCSI : SUCCESS in 3h 01m 11s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/5/20230118083943/NEC-V-FC : SUCCESS in 6h 22m 53s","accounts_in_message":[],"_revision_number":5},{"id":"c544d84b827af8c31b1fb028c55b123fcdd187f4","author":{"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},"date":"2023-01-19 06:50:17.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5453c1a0c30a3943218cf945e48c9b9f707cba6b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-01-21 06:16:41.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/5/20230121083500/NEC-ISCSI : SUCCESS in 3h 33m 26s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/5/20230121083500/NEC-FC : SUCCESS in 6h 41m 38s","accounts_in_message":[],"_revision_number":5},{"id":"05c185736b76697dc257794b29172d54972dcd89","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-01-24 10:27:55.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b239bbfd479c416c3cb46c2f0257f3de9aa98439","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-01-26 12:35:12.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"23cb827ab3639f235116ffc1bbc5536ebb0d9dbf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-01-26 12:35:21.000000000","message":"Patch Set 6:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":6},{"id":"979b373f6e3c0f8ec7dbdd5a7763625f98fbe76f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-26 12:36:53.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"760664cdb2b8f57c4540fe58aa959cbcd81474ec","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-26 12:46:56.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"d83791512d39f1f1c24aceceeab73aa180715b78","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-01-26 12:52:43.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/cf1ee7475f954deb9e89d55c1d05358e : POST_FAILURE in 7m 45s","accounts_in_message":[],"_revision_number":6},{"id":"898fe3d1307f56d8708c53595101a9820cf6ecd5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 12:53:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d50d50ff922488cbcc17b9225209dbd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/17f5e8bb59c9486c9400323b86535e3b : SUCCESS in 12m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a23850af68cc4615bb8e19bf58e1c5e8 : SUCCESS in 16m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8f40fecf4f14e34242d1155ea332bf2621fac15b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-26 13:12:50.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/847132 : FAILURE in 23m 38s","accounts_in_message":[],"_revision_number":6},{"id":"651f7127e4e0d82351be29dc6d8de4a641aa953d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-26 13:24:51.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/847132/6/check/iscsi-driver-alletra-6k/0bb5b28/0bb5b28.html : FAILURE in 25m 07s","accounts_in_message":[],"_revision_number":6},{"id":"5b01a0365f80d4384eb7986eed64d0bb69f398fc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-26 13:41:01.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847132 : FAILURE in 26m 22s","accounts_in_message":[],"_revision_number":6},{"id":"a4ee3aa3a7cb859e57d1b01a87721ee1c2ec2d0e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-26 13:47:52.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/114288dc54fc465f83b3d422b6e784bd : SUCCESS in 1h 08m 00s","accounts_in_message":[],"_revision_number":6},{"id":"fe3be4ba37687127f41bb180daf1f4ba88be1ddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 14:42:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65c06218628f4059b93327c5a341ccce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47b3142df0274071806630117c220aa7 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d90c66c5da8547f38da216abd6ff2db4 : SUCCESS in 9m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a61ffde4cb344e4b0006c345fb3cab2 : SUCCESS in 9m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6adab5066aa4458a9425ba5396364443 : SUCCESS in 10m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/615090dac4ae45809858b2d648d1de9b : SUCCESS in 1h 05m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/50d9fd565c564b3c9826097b97ff18af : SUCCESS in 53m 01s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c5cdde258512426caaab879e2433cffd : SUCCESS in 1h 16m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a810c7b9d1e349698b5b80a9a0655c45 : SUCCESS in 47m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/60ff178c6737465cb7165275da62b0b6 : SUCCESS in 14m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b5a774f789f64186a520a3c0c20222d1 : SUCCESS in 4m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c73b38d838584f12800e93bb2cffd897 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ff8e48c16cda4606924dd77610de6b64 : SUCCESS in 5m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c9981c78141a476199f0b7d1924b604e : SUCCESS in 8m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0617cf5ceb924987a618188789fd7969 : SUCCESS in 34m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dc7121f3166241be9a6eb3c4cd249afd : FAILURE in 4m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8406835e479d449ca76e102645d5fb69 : SUCCESS in 1h 29m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1265eb93416943c682374b20e3bb6480 : FAILURE in 1h 33m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f64d084498c440028fa2672489666f26 : SUCCESS in 1h 32m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/269d96f57a4143a2afd635663b120c03 : SUCCESS in 1h 09m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7d10ed85d2514e58ba11d660606894f0 : SUCCESS in 1h 25m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/871d452fa0e248f0a0fac2895be6856f : SUCCESS in 30m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5c1fc5eb291f41a6baed8a496827e6cd : SUCCESS in 1h 43m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/68642dcd772b4feca132ea2fac2413ea : SUCCESS in 2h 00m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/602fe133f28d40caa64e5f61411ddc98 : SUCCESS in 49m 22s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/320dbd72ff8a4c5ea4f65f96ec940a69 : SUCCESS in 1h 32m 33s","accounts_in_message":[],"_revision_number":6},{"id":"61898e04f21c14dc7815329644a2ba5d47cbd082","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-26 14:50:12.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/26_01_2023_14_42_38\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"9d5f26c19f10f34c4be6dd6ea65a14a08143c1eb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-26 15:41:26.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/6/upstream-check/cinder-tempest-plugin-ontap-fcp/7af63db/ : SUCCESS in 57m 07s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/84147dc/ : FAILURE in 48m 46s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/6/upstream-check/cinder-tempest-plugin-ontap-nfs/9fcb0ad/ : SUCCESS in 44m 02s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/31958e9/ : FAILURE in 31m 54s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"1f368b63a9299059a7a31e92ff5ac1cb17905512","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-01-26 16:05:03.000000000","message":"Patch Set 6:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/32/847132/6/fc : FAILURE in h m 0s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/32/847132/6/iscsi : FAILURE in h m 0s","accounts_in_message":[],"_revision_number":6},{"id":"fda31979739d5067b2ec0c29779a0536c19fb069","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-26 23:15:28.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/32/847132/6/ : FAILURE in 1h 11m 03s","accounts_in_message":[],"_revision_number":6},{"id":"99172a463bd68adf00e3d8be26a24274aa35c15d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-27 00:08:42.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-847132-6/ns5_nfs : FAILURE in 11h 7m 40s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-847132-6/ns5_iscsi : FAILURE in 11h 10m 29s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"a013fa43d5a0eb01ce56845bd3ac776b530d033e","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-01-27 08:13:25.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/847132-6/HPMSA-iSCSI/ : SUCCESS in 1:36:49\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/847132-6/HPMSA-FC/ : SUCCESS in 1:17:35","accounts_in_message":[],"_revision_number":6},{"id":"a64fa1041f8a405fd484f6db04326ceb5d3940b9","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-01-29 03:44:54.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/847132-6/PVME-iSCSI/ : SUCCESS in 1:52:57\n- PVME-FC https://os-logs.seagate.com/dell-ci/847132-6/PVME-FC/ : SUCCESS in 1:35:19","accounts_in_message":[],"_revision_number":6},{"id":"34fd8a4633faec9f744600006e6f0a227c41ee19","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-01-29 03:46:15.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/847132-6/Seagate-iSCSI/ : SUCCESS in 1:51:13\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/847132-6/Seagate-FC/ : SUCCESS in 1:34:58","accounts_in_message":[],"_revision_number":6},{"id":"0c54fbe2cb37a9795c6ba4fb5ecaa141b1626f85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-02-08 13:33:35.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"7869f9919e42cb81fdf7fa25f8553be95ff350f2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-08 13:35:56.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"45d3601598bb97113354a514d5cee5fdc767cc3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 13:49:28.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e0733c159c941ccb1235aeece449d79\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2375a757ab9243229287086cbe9fbdad : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30f93e13f829412b9b52ebdc385e46c5 : SUCCESS in 12m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5e3e60fe2bb62786619244b31ff1fe21a3489eb5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-08 14:12:34.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/847132/7/check/iscsi-driver-alletra-6k/f154c97/f154c97.html : FAILURE in 26m 15s","accounts_in_message":[],"_revision_number":7},{"id":"1ad652cafaf6fe9ed666205d50ddbb8fecc0b259","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-08 14:14:29.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/847132 : FAILURE in 23m 30s","accounts_in_message":[],"_revision_number":7},{"id":"d556cfe371a56a57b07d4d4aeb1fa73ae927129d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-08 14:18:50.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847132 : FAILURE in 26m 09s","accounts_in_message":[],"_revision_number":7},{"id":"9512494df5f8781a3ffd895a6f1f27f63decee33","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-08 14:38:41.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/32/847132/7/ : FAILURE in 56m 49s","accounts_in_message":[],"_revision_number":7},{"id":"36b47799041761738a5b4b82847a0ee7156fbfc2","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-08 14:40:20.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/c9d70c5979734dcc881b18077ff7dedc : POST_FAILURE in 35m 08s","accounts_in_message":[],"_revision_number":7},{"id":"bdac4ed6ab1aca98ac89767d41b29aa603dd4db3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-08 15:10:41.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7cef18039e3b4fa3b66d462798545639 : SUCCESS in 1h 31m 35s","accounts_in_message":[],"_revision_number":7},{"id":"d8a5cfae98d7163d8d7028d79aa871293bb5acbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 16:29:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b929dd0ddff476b8778c835e3bf64b9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1afe99e4caac4e0d8892b3d2ba9fe5dc : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35b37161082d473c94997e5fa44b4e3b : SUCCESS in 11m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fcedf6accd0c4f36b2cc9c4a27f71e99 : SUCCESS in 10m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/454c42585f1a489badaff97723d5559e : FAILURE in 7m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/6b5a69bcff2c4e0b9b3f1afc9fb3022e : SUCCESS in 58m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/96f0b99825834ebc8e21f961c6c83147 : SUCCESS in 1h 16m 40s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bf13d6a265174f8d9c08f3df1557d6ad : SUCCESS in 1h 35m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a49838229fee4c61b59d65f8975b29b7 : SUCCESS in 1h 03m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3dec6ea8458e4a568f6075a105efb1e1 : SUCCESS in 15m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e72c2e29635c4e7c876817ebf9107142 : SUCCESS in 5m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fa19dd8de4f44049b19a57c48f774d55 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3d4c36d5fe20483fb0d2e9e978367e01 : SUCCESS in 5m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/882639f3835f46f395285bb002c9ff4c : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2399b9132eb742e09ffe3c9136348c2e : SUCCESS in 52m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8f210eb4562f49019a3e3fcb58cc01c0 : FAILURE in 6m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2db95d7feabc4211945c75b8fc2bb9c9 : SUCCESS in 1h 12m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/525509b0c90343c1b2d4ca83b20c4893 : FAILURE in 1h 26m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/463d7accf1ad4864bd3ae0c9f7b86274 : SUCCESS in 1h 46m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/060e5397518b4aec98a09b38e8d5a198 : SUCCESS in 1h 15m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/79363e6ca5494b22b369b1d6f2eb5f9c : SUCCESS in 1h 35m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/120c1ab220d9414e9f14faad30efb460 : SUCCESS in 1h 26m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d0e32e7d22584fec9d57e7b428754338 : SUCCESS in 1h 30m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e6e2d574d1044118bf753b2602690a04 : SUCCESS in 2h 06m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9556a0e940814ded9f6bbe38bf570372 : SUCCESS in 51m 10s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98eee26b7ab74c939c3b6941d177dd91 : SUCCESS in 1h 33m 13s","accounts_in_message":[],"_revision_number":7},{"id":"3b0f46ad207362b9f11ba01ee16b989c6e643e0c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-08 16:29:59.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/32/847132/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2dd2b11/ : SUCCESS in 1h 47m 59s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/32/847132/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f980c39/ : SUCCESS in 1h 56m 50s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/32/847132/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/95c1478/ : SUCCESS in 1h 48m 43s","accounts_in_message":[],"_revision_number":7},{"id":"0079a0d421541dc26c03dbab467ea8595a55740b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-09 01:34:14.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/fea791fa4f904f42880270c2b11fa5ec : SUCCESS in 2h 03m 07s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/d0d3ef4b923e4434856e4b77af7e9543 : FAILURE in 1h 32m 01s","accounts_in_message":[],"_revision_number":7},{"id":"f76da6e6b03076b8ded83aa5012af73bceac6acd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-09 01:50:52.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-847132-7/ns5_nfs : FAILURE in 1h 7m 14s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-847132-7/ns5_iscsi : FAILURE in 11h 54m 12s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"51d97579cbf0c1dc6decedf12e1955ce4c39fa30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-02-13 09:20:32.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"37465a50328e4266e4d5d57c41d70754bceca19c","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-02-13 09:21:55.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"426e6cc7abd14a4a06f9bcce920214af707c0d61","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-13 09:24:44.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"f2e5cb8f3b4887c6ac7213bc6e98ecdddde675ac","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-13 09:49:44.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/847132 : FAILURE in 23m 34s","accounts_in_message":[],"_revision_number":8},{"id":"aceba57e8bb5cfb2287df1daa77d0d4539c0a95d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 09:50:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a704e983f2864291b5446c6da4bc23c1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8a52516638041898f030c9bc728c8b2 : SUCCESS in 22m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3d1c21590c0645b892fadd0c656e0368 : SUCCESS in 26m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6a4f3f356a613b521feaa429f35e3736ba89a311","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-13 09:50:33.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/847132/8/check/iscsi-driver-alletra-6k/09e3f87/09e3f87.html : FAILURE in 26m 11s","accounts_in_message":[],"_revision_number":8},{"id":"df687b041915ea1c7d17d04b98b2612a98f64303","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-13 09:50:48.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847132 : FAILURE in 26m 17s","accounts_in_message":[],"_revision_number":8},{"id":"246b2058011442d07bf4e5c8817944cfeb8f14d1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-13 11:33:16.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/134b89fd7eb640a88fc82ba7b66da4a1 : SUCCESS in 1h 16m 40s","accounts_in_message":[],"_revision_number":8},{"id":"dfebcd49e20373867c297a0282ff92600cd74c30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 11:46:13.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/282c90fdc0664ff2a10da5d161a3c687\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c02d5ff44f744c429d1459e7140d5b55 : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83f1f7cff56e481b834a9297795f03d4 : SUCCESS in 12m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1223d27830c9430fba67c1589b0ab589 : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01d2252085b849cc99b9b8fc0a8642d0 : SUCCESS in 13m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/3a655827203a4a61ab98a03859b91317 : SUCCESS in 1h 08m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/43582cb474df415ab23ad557c9f5f4cc : SUCCESS in 1h 12m 13s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4125a899c56414ca216aebc23399985 : FAILURE in 1h 21m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fddae46c61a84e7d8c8a2c46f925486f : SUCCESS in 55m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5087c02f51dc47489074ba1935f7163d : SUCCESS in 15m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/616359999aa14714ac5862543ce109de : SUCCESS in 4m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/366fcbd46c0e4da9a5613eb9704d4e08 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dc068654c0464da7b7422e1bcdedba46 : SUCCESS in 5m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/64701756c8ab4c2fa5bb8223d33f5cc1 : SUCCESS in 6m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4219e529afcf41348548561fb97e68d7 : SUCCESS in 46m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c2c469b8fe4f4d0fb934609066842deb : FAILURE in 5m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3b7ca6a6c61c4f61bfc8247babb16869 : SUCCESS in 1h 09m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b98b6ba2e5ed4002bff2d690e520aae9 : FAILURE in 1h 23m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/17e298fc207c431690ac7f3a278a71cc : SUCCESS in 1h 47m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e37c3ecabea04ffbba42995cdf15c0cb : SUCCESS in 57m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3f97e3be2c3a481caf704159d68d9a20 : SUCCESS in 1h 18m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a809db070cbf49cea12736f4f4fe0c40 : SUCCESS in 38m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a0d76f601121440e832a923bb5926229 : SUCCESS in 1h 25m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/395683d49897445bb1fdf22965c4d05d : SUCCESS in 2h 17m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6006e8cb5133471bb3ec7da09a5b31b1 : SUCCESS in 49m 27s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac0019dc2bbd47daa737042f1a73016c : SUCCESS in 1h 08m 35s","accounts_in_message":[],"_revision_number":8},{"id":"d268de6ed9a3234de742ba0bee3afc7856d4aba2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-13 14:49:51.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/32/847132/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/fdc8003/ : SUCCESS in 1h 52m 50s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/32/847132/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/79ee528/ : FAILURE in 1h 52m 20s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/32/847132/8/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/52ab05a/ : SUCCESS in 1h 41m 15s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/32/847132/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/01ca8b4/ : TIMED_OUT in 3h 07m 15s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy pure-devstack-cinder-tempest-nvme-roce-aio-jammy : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0ad95ab5adbcca2502e7054764a2f79e9d175e27","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-13 15:56:09.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/d3bc4de1b2e84c11ba71ddb61adf0e3a : POST_FAILURE in 47m 50s","accounts_in_message":[],"_revision_number":8},{"id":"a1415ce557a73a99c2875e9c07f7f73e649af835","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-13 18:47:31.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/847132/8/check/DellEMC_Unity_FC/6eedb60/DellEMC_Unity_FC/2258 : SUCCESS in 1h 24m 03s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/8/check/DellEMC_Unity_iSCSI/f47fef4/DellEMC_Unity_iSCSI/2118 : SUCCESS in 1h 49m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"d765902a82d6aa733b89d166ff5f0ae55a8ddf6f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-13 19:00:29.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/d748156b35b34aa7b32ab890cf9c5b25 : SUCCESS in 2h 20m 57s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/b08cacd2244f4173a5ea4108b450af91 : FAILURE in 1h 37m 15s","accounts_in_message":[],"_revision_number":8},{"id":"ca5ea5b56ee7c76e57e30b2f3269c34a337eeb1d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-13 21:45:46.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/32/847132/8/ : FAILURE in 1h 13m 53s","accounts_in_message":[],"_revision_number":8},{"id":"8fb7e6b03420fc70816b7ea76edb211f63d05062","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-02-13 23:51:43.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9b73c03dedb87bef96927228a0b85ff169c13928","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 00:06:13.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de1a2a8bfaeb49b497d3de7c8eed9310\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b5a114ab206476ba98181bdd64173b4 : SUCCESS in 9m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ee66109f82564673a532ea890580d0e3 : SUCCESS in 13m 19s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"025b01fedc9f452f089da8b9b6568d4161fcd690","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-14 00:22:02.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-847132-8/ns5_nfs : FAILURE in 13h 48m 6s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-847132-8/ns5_iscsi : FAILURE in 13h 46m 22s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"9eb51090641dc50e23765180b651ed2f7615ac49","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-14 01:14:07.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/af3e06bacaf34ca591ad27e46dbb6b3e : SUCCESS in 1h 20m 34s","accounts_in_message":[],"_revision_number":8},{"id":"75e0295eb2b5fd1fb6e291c513ca72bec49ac4e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 01:40:43.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa6ef96a072c458db057d433445519b2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf86c396cf34020839492e92e9bdea3 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b7084b381014f5c82d368c0661cdc87 : SUCCESS in 11m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1612b3dd8d1848f4aad1224bad3d8d08 : SUCCESS in 10m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00aefeec65634c1792fc9f778d50d175 : SUCCESS in 11m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/98b3832c7d4647849b8e8fa92c7de589 : SUCCESS in 1h 03m 13s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ac4d30874ea7497dab9a461b70078f08 : SUCCESS in 44m 30s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b990da66cf494b06b5d322c1a16d86ce : SUCCESS in 1h 09m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf5411a112c14360b1c12207c371d4b0 : SUCCESS in 1h 00m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6a59285f578242da96641a472f6746b3 : SUCCESS in 13m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bd41173983de406c808e7120f9896381 : SUCCESS in 4m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1f421d9f009d4cf4babc0ec1571eaa53 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fc2cc3fcd3b1423e86029114e6e984c9 : SUCCESS in 5m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6047398f11ed47879c186cf90531249b : SUCCESS in 5m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/520251fb490e421d8682976902adccf9 : SUCCESS in 49m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ad171586be7d4ac4a5fab875b49f6071 : FAILURE in 6m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1e345c747d384d17b15cb0a2eec6039f : FAILURE in 1h 20m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e8ea217a347442679697e273fca5ed34 : FAILURE in 1h 15m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a164d85d1d6a43b3818ac926ecab0e2a : SUCCESS in 1h 42m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/878df34a9ecc4cb597715c7f03317cc5 : SUCCESS in 1h 15m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f6f914652f694f56afba7bd97cfabd88 : SUCCESS in 1h 20m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0f13ac16a55c40f683464d4d8a2a75aa : SUCCESS in 1h 16m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/285f6513bcc240da95634bdc3ce9a908 : SUCCESS in 1h 30m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/add718a78fa14569aed1c103138b0763 : SUCCESS in 1h 46m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc45923b987044499c20fba787e0cd77 : SUCCESS in 49m 34s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6659fdecf9f9469aa0709161916b9bf1 : SUCCESS in 52m 58s","accounts_in_message":[],"_revision_number":8},{"id":"43deb16d8b139c5c9e7b4eea95aebfc30259c9b6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-14 02:22:49.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/32/847132/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d279102/ : SUCCESS in 1h 36m 29s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/32/847132/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/4369dd5/ : FAILURE in 2h 07m 02s\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/32/847132/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/788ae19/ : SUCCESS in 1h 52m 10s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/32/847132/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/ac5666e/ : FAILURE in 2h 29m 55s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/32/847132/8/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/e6481a7/ : SUCCESS in 1h 54m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3fef2cf5cfe99d4c2da78a847e8394c996e7a018","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-02-14 06:26:06.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b9334f243cac3e3b8f93c62008ac41c6657ab84a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 06:37:25.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9228dbe408cf4ad3a624fef91f348dd3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec3d8d4cc229402f8c25c2a21066bf5e : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1c15b5863f5c4e9db55bc75e2b0e0abe : SUCCESS in 10m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6a98d30c06ae3430d22a91b508a4b8be93341b49","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-14 07:35:15.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bffa8678450f4166b7e45bd440dda88d : SUCCESS in 1h 07m 19s","accounts_in_message":[],"_revision_number":8},{"id":"17c145d906f178689c00c84287c109fc830406dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 08:25:29.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ce44c7f83274332ac7f5edb80fa49b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56bf0c4c25e04f24baa18627e23cad13 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5601cfe25b2d4c3b85b81f2fdd646e96 : SUCCESS in 10m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6eddda2c6ac74ca0a469a7b60c352e82 : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f752f8f082e42aaa3d4980fb6a4162a : SUCCESS in 13m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/069d04c37c0e4bbf9f368baf948137ae : SUCCESS in 46m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bff8190186fe402385e666227af64ba5 : SUCCESS in 41m 16s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b5ca9088da845d7b08361529695f8be : SUCCESS in 1h 04m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b426b597e974aab8cf353b4a5771a66 : SUCCESS in 38m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/93de8096e0dd4992b1cdb291c69b8901 : SUCCESS in 14m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cd5a7a6454874c43a865ec2de5485223 : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/08b6e2940c54436c821cc11e5e1a4e3c : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/69bfd86739d6474cb31335550221d5dd : SUCCESS in 5m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ccc5e4e83a8941c99cd121f6b6d0a8dd : SUCCESS in 5m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a08d4b3e5d26483689eaf0cb4ea30dae : SUCCESS in 33m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/116a6c899b7347a99ab549219d0e6b49 : FAILURE in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/74e3d83e508a498c802aa96f8c8b55f3 : SUCCESS in 1h 19m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6585b4f67e23418b8fb82be18482ef57 : FAILURE in 1h 06m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7590abbcbdc545eba8e457f2e0b2927d : SUCCESS in 1h 58m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/db100545f21c42ee904a8d1cf2ea568f : SUCCESS in 57m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6cef8b9a0a2b46b480e88da19838652e : SUCCESS in 1h 16m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cbb72856ec534dc7998021259d656cb6 : SUCCESS in 30m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/842a1d6df38b47f3b77d5054ea26d0a8 : SUCCESS in 1h 07m 12s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2e44f4ca1e264e02bacaa3e06f0212f8 : SUCCESS in 1h 56m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab0d97fe48d141f7b7b0ec5a512f1fd4 : SUCCESS in 41m 41s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08505797222847589c4392d40fafe827 : SUCCESS in 57m 56s","accounts_in_message":[],"_revision_number":8},{"id":"ca0aa2bea32b524f1085d81f62e09b142554971e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-14 08:57:00.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/32/847132/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d45275a/ : SUCCESS in 1h 43m 45s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/32/847132/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/24b414f/ : SUCCESS in 1h 37m 21s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/32/847132/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/32137f4/ : FAILURE in 2h 23m 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/32/847132/8/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/f26a3bc/ : SUCCESS in 1h 45m 12s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"730c135f923264b2a7805bc5cf5dec00815c307f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-14 09:31:00.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/8/upstream-check/cinder-tempest-plugin-ontap-fcp/0c60deb/ : RETRY_LIMIT in 8m 33s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/2dc6337/ : SUCCESS in 1h 04m 44s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/8/upstream-check/cinder-tempest-plugin-ontap-nfs/e186c9f/ : SUCCESS in 43m 42s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/8/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/e8b3c57/ : FAILURE in 42m 55s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"75862427665afc26a8e70150a48e234c37a1588a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-02-14 09:53:36.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/32/847132/8/check/DellEMC_PowerFlex/23367ec/DellEMC_PowerFlex/1606 : SUCCESS in 1h 27m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"ebfd66c228ce602f2bfb4c4ebd9078c6acb7675f","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-14 10:47:06.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/a8d9e4382d744fcf8f9bdfdf2284d627 : SUCCESS in 1h 11m 28s","accounts_in_message":[],"_revision_number":8},{"id":"43062544555ceeef65c4fcc59e26bb4ef4bf3b33","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-02-14 11:05:03.000000000","message":"Patch Set 8:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/32/847132/8/fc : FAILURE in 1h 48m 04s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/32/847132/8/iscsi : FAILURE in 1h 02m 07s","accounts_in_message":[],"_revision_number":8},{"id":"b77902960deab6e610e6e188b419a17cf5157426","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-15 00:35:59.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/847132/8/check/DellEMC_Unity_FC/158dced/DellEMC_Unity_FC/2273 : SUCCESS in 1h 18m 45s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/8/check/DellEMC_Unity_iSCSI/6a44e68/DellEMC_Unity_iSCSI/2133 : SUCCESS in 1h 45m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"08d1a5e468f11080f00d58ed4c503e5d20faf8cf","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-02-15 19:50:34.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/847132-8/HPMSA-iSCSI/ : FAILURE in 1:53:00\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/847132-8/HPMSA-FC/ : SUCCESS in 1:16:38","accounts_in_message":[],"_revision_number":8},{"id":"d80e172b4de883de59eb857644b1d5f19a19269c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-15 22:58:50.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/6d4111c325f7439f8fd562cdc8b5bdea : SUCCESS in 2h 01m 56s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/d049e78561b74213a6737e9756cddef0 : RETRY_LIMIT in 1m 00s","accounts_in_message":[],"_revision_number":8},{"id":"b6f5e9d109d77f5439e51c0baa5bbc6bda08dfd1","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-02-16 03:59:59.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/847132-8/Seagate-iSCSI/ : SUCCESS in 1:54:07\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/847132-8/Seagate-FC/ : SUCCESS in 1:36:21","accounts_in_message":[],"_revision_number":8},{"id":"f8d8f0f0213288c5f8c6bd76c7f8bfba5f2ba27f","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-02-16 04:09:13.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/847132-8/PVME-iSCSI/ : SUCCESS in 1:55:21\n- PVME-FC https://os-logs.seagate.com/dell-ci/847132-8/PVME-FC/ : SUCCESS in 1:35:34","accounts_in_message":[],"_revision_number":8},{"id":"ce3c9a777bef03a0bc7c151ac3d2933b5290edd9","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-19 00:42:07.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/8/20230219071445/NEC-V-FC : SUCCESS in 2h 15m 49s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/8/20230219071445/NEC-V-ISCSI : SUCCESS in 2h 27m 19s","accounts_in_message":[],"_revision_number":8},{"id":"dde5929ef12245309e2a140eb19cb9a47c93f6d2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-02-23 13:32:18.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/8/20230223184704/NEC-FC : SUCCESS in 3h 27m 49s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/8/20230223184704/NEC-ISCSI : SUCCESS in 3h 45m 11s","accounts_in_message":[],"_revision_number":8},{"id":"a93e8d4e829e73c39bb90ef488daba545feb4817","author":{"_account_id":32425,"name":"saikrishnagk","email":"gopalakrishna.krishna@ibm.com","username":"gksk"},"date":"2023-02-23 13:46:18.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"8a507b01c50f91bccfdac88b55c28108d5d2a760","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-02-28 10:29:54.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"36ec33b144b4c141370116afe75dd2999141bf53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-03-01 12:49:08.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"191d78bc89ff384d5521b10021fde4379294ff07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 13:25:38.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/690d929bb6bb433895ddfed9b891cce7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/582b07b900564634acf68d8d8824300c : FAILURE in 33m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/575c11785c8f47ffa9751dc25437d11b : SUCCESS in 16m 38s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8f84010b592ed1a128ef95918bce5210eabe8dd3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-01 13:30:02.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/847132/9/check/iscsi-driver-alletra-6k/efcd044/efcd044.html : FAILURE in 26m 14s","accounts_in_message":[],"_revision_number":9},{"id":"ccfd293b7b673267d43984871ddcd32bcd8450a8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-01 13:32:23.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847132 : FAILURE in 26m 06s","accounts_in_message":[],"_revision_number":9},{"id":"36fc9722ac78094c4c7b938c4d6bdd24f0472b3d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-01 14:02:24.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/966f8323074d4aa9a1e1bd98fbe07914 : SUCCESS in 1h 07m 01s","accounts_in_message":[],"_revision_number":9},{"id":"222171dadee2281942e5ef3aa24b019421cdeb96","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-03-01 14:33:16.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/32/847132/9/ : FAILURE in 54m 18s","accounts_in_message":[],"_revision_number":9},{"id":"adaf48aae90664e81b7cdb50dd74b89988918727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 15:28:17.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8204db126dff47f7bffcfa75234fe715\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1c14271997d475295e1b5df60bc1788 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29a37f5b912f4eabae17161b3f78fc14 : SUCCESS in 10m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/189c36a923a34332a0958b3a27279e5b : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c461965aec8743ffb9006f6a98b897a6 : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/d336d4cc38ad4689b43fe40cf88e705f : SUCCESS in 51m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9097e9db05734ef2a66499f4f374de04 : SUCCESS in 1h 02m 21s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b0dd2a52c5cf4dc3b5c211f3bddc3776 : SUCCESS in 1h 33m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02239d0958f04ff6bd606b51d388c912 : SUCCESS in 1h 05m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dfc2d419ff064009aba35194088823e5 : SUCCESS in 19m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3cb41676a9564f129b5bb86c08176a47 : SUCCESS in 6m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4a2387c9351841be9361969bd56fa77d : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6e5cea139e2d40c3800bd11f8111b30d : SUCCESS in 7m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6d6283ead8be4d8982056a0c799206f0 : SUCCESS in 6m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a0a0616d58514d4e8aea093b6e8fbe89 : SUCCESS in 50m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6ea20e9ac5524eeb8504c09bca2e8e57 : FAILURE in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/15d324cb8e544d5f881f7168695f92d7 : SUCCESS in 1h 14m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a48fcae79f4b49b39bcaf2a31d9ce1de : FAILURE in 1h 39m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1c828b78018545809e252a9a451a7fc5 : SUCCESS in 1h 49m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/361ba90421ff4dca99fc7959a874bb1e : SUCCESS in 1h 49m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fdb7dfce5aab48f9ba078d98092560c4 : SUCCESS in 1h 34m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0013e747d2e94768a21fd66ff51d78cb : SUCCESS in 47m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/973bdc505c7d43b0a5fb86baed0eda93 : SUCCESS in 2h 30m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/372159f1124d4475b383d30717bf2902 : SUCCESS in 2h 12m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bccdb4768c9a4666ab7e39b3ad11a027 : SUCCESS in 56m 08s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ec4620ba8be4c289c800241537b1ede : SUCCESS in 1h 22m 22s","accounts_in_message":[],"_revision_number":9},{"id":"66a8ea976a69e4044d0ab84f54cc5deb71f8351e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-01 15:40:05.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/847132 : FAILURE in 23m 46s","accounts_in_message":[],"_revision_number":9},{"id":"f7a75011ee53cc5f00dcdfcf1e76439d38875fb5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-03-01 15:43:56.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/01_03_2023_15_36_07\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"0e0681004c07107df0494166cc865cea10bf2575","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-01 18:09:10.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-847132-9/ns5_nfs : FAILURE in 4h 19m 19s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-847132-9/ns5_iscsi : FAILURE in 4h 51m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"942df2d48fdccfc91b9fb3a8a2517559e4a01852","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-03-01 18:17:07.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/9/upstream-check/cinder-tempest-plugin-ontap-fcp/ade570f/ : SUCCESS in 2h 47m 33s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/9/upstream-check/cinder-tempest-plugin-ontap-iscsi/248e284/ : RETRY_LIMIT in 7m 50s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/9/upstream-check/cinder-tempest-plugin-ontap-nfs/11bf718/ : SUCCESS in 42m 18s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/9/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/5ed230c/ : RETRY_LIMIT in 7m 29s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"e2e0adc9ee6a8c1396514b38e8b7cb443543b9a2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-01 18:27:19.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/32/847132/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f5a5e81/ : SUCCESS in 1h 55m 39s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/32/847132/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d8c8cb3/ : FAILURE in 2h 50m 15s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/32/847132/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/bb597ef/ : SUCCESS in 2h 08m 32s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/32/847132/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/3f4bd00/ : SUCCESS in 1h 47m 04s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"435e662c810ad9c6103e4fbcdb98073c80802820","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-03-01 21:38:35.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/9/check/DellEMC_PowerStore_iSCSI/04b1db3/DellEMC_PowerStore_iSCSI/1068 : SUCCESS in 1h 12m 56s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/32/847132/9/check/DellEMC_PowerStore_FC/3785066/DellEMC_PowerStore_FC/1191 : SUCCESS in 1h 30m 41s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/32/847132/9/check/DellEMC_PowerStore_NFS/d9beab3/DellEMC_PowerStore_NFS/707 : SUCCESS in 1h 22m 50s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/32/847132/9/check/DellEMC_PowerStore_NVMeTCP/542fa97/DellEMC_PowerStore_NVMeTCP/572 : SUCCESS in 1h 38m 41s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"86c93fb8753f90c2621503d15c36f0c4b6a6bf59","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-03-01 22:08:54.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/32/847132/9/check/DellEMC_PowerFlex/336a91d/DellEMC_PowerFlex/1674 : SUCCESS in 1h 17m 37s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"0b01d182e6e3d1cbc5388cf3112d977513c4f0cf","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2023-03-01 22:19:35.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/32/847132/9/check/DellEMC_SC_FC/b4ed9ab/DellEMC_SC_FC/1139 : SUCCESS in 1h 06m 50s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/9/check/DellEMC_SC_iSCSI/83c2f8e/DellEMC_SC_iSCSI/1123 : FAILURE in 32m 16s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":9},{"id":"aab20cb4fa035d94fdc7a48ac74f7b3e0a5f5896","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-02 00:04:15.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/dddcaa7f773c4c14b092ed38991ea102 : SUCCESS in 2h 11m 24s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/6c17f06ba804449390912055fa68a9f9 : FAILURE in 1h 49m 15s","accounts_in_message":[],"_revision_number":9},{"id":"6dfa774d22c98a4973695051d529d92ea61274bd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-02 01:04:36.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/847132/9/check/DellEMC_Unity_FC/89b412d/DellEMC_Unity_FC/2393 : SUCCESS in 1h 32m 21s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/9/check/DellEMC_Unity_iSCSI/424f8d0/DellEMC_Unity_iSCSI/2245 : SUCCESS in 1h 57m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"6d6fe88df36a82053852ed8bbe187eaada82e66b","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-03-02 20:48:46.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/847132-9/HPMSA-iSCSI/ : SUCCESS in 1:37:43\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/847132-9/HPMSA-FC/ : SUCCESS in 1:16:30","accounts_in_message":[],"_revision_number":9},{"id":"8cb7d9911d0fcfa14257e2d1188377db97fc69d9","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-03-02 23:05:43.000000000","message":"Patch Set 9:\n\nBuild failed.","accounts_in_message":[],"_revision_number":9},{"id":"72a30db6a49576b9c3338b6d236dc8305e6c8d3e","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-03-03 02:49:14.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/847132-9/Seagate-iSCSI/ : SUCCESS in 1:49:56\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/847132-9/Seagate-FC/ : SUCCESS in 1:35:09","accounts_in_message":[],"_revision_number":9},{"id":"9b94cece14fe3be923b7f1a019ed753144972498","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-03-05 22:20:45.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/9/20230306012117/NEC-V-ISCSI : SUCCESS in 2h 28m 09s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/9/20230306012117/NEC-V-FC : SUCCESS in 5h 59m 25s","accounts_in_message":[],"_revision_number":9},{"id":"56f958dfb8d3dfb2f809543e30faf05faefd3822","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-03-07 05:03:10.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/9/20230307103014/NEC-FC : SUCCESS in 3h 29m 19s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/9/20230307103014/NEC-ISCSI : SUCCESS in 3h 32m 53s","accounts_in_message":[],"_revision_number":9},{"id":"6de269c61565067b99927b60134c279c52957607","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2023-03-08 22:23:27.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b6eee5640eede9085365fb4c94b2c5aff87616f5","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-03-09 23:57:43.000000000","message":"Patch Set 9:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/32/847132/9/fc : FAILURE in 0h 52m 11s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/32/847132/9/iscsi : FAILURE in 0h 43m 34s","accounts_in_message":[],"_revision_number":9},{"id":"7fd4f8b865d49bf213e8566acec3f15598856de1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-23 00:19:29.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/32/847132/9/check/iscsi-driver-nimble/a0a7219/a0a7219.html : FAILURE in 12m 31s","accounts_in_message":[],"_revision_number":9},{"id":"7202ed67cabf500ec51d6943e711a30d989388cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-05-08 19:32:02.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"da24447918e0ec3f67edb2cca3ea8a937613609d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 20:00:20.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/449e6954a593498e8def214d8f5fab81\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/618904f00e45405fa02d313323a7b018 : SUCCESS in 25m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a8c6c4660784437d8b9ba95e885235e2 : SUCCESS in 12m 16s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"55b3f1ba60c6eb50941e6b9c7dde618407b2ac3e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-05-08 20:37:18.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847132 : FAILURE in 1h 00m 12s","accounts_in_message":[],"_revision_number":10},{"id":"a9d10b77db6c7b81010a626fbffc1bd7a0be39ff","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-08 20:39:45.000000000","message":"Patch Set 10:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/a7caf1a729674c778bbcb2a88b7d5fd0\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b1cabf6ab7d14eb2b1718a0c896df509 : FAILURE in 50m 46s","accounts_in_message":[],"_revision_number":10},{"id":"a3d23172e15195f7ccb07360c870be0240c6d9eb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-08 20:42:00.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/847132/10/check/iscsi-driver-alletra-6k/ef2543e/ef2543e.html : FAILURE in 1h 07m 46s","accounts_in_message":[],"_revision_number":10},{"id":"5832599c7e6cdc7462f87fd7399e35b87d31d216","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-08 21:27:27.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/e4efb1c9750748e6b628e136e2e24d08 : POST_FAILURE in 54m 51s","accounts_in_message":[],"_revision_number":10},{"id":"226105bec49ced6bc02d385be4dd909351ee3b56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 22:14:42.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/709fae4182484a7abc78f9d9e01550d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d232af2acde41f3a6d494aa4b1488ff : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f7b60ab9d4743cc8333fe174601ed7a : SUCCESS in 12m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/858a2a262f57406abb667572ef133452 : SUCCESS in 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f4b125396b4d6d9bf04ecc9ecddd59 : SUCCESS in 12m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/e3c42b4262ed4852834e968ab3b90ac0 : SUCCESS in 56m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3d87c081b8464588a3b4aa33271727b3 : SUCCESS in 1h 16m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3284991eff1f4143ba756f3eac1b35ee : SUCCESS in 58m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da60558dda5a482fa95cf3790ad15767 : SUCCESS in 7m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/964568e89e9f410797311ef4cfa1f502 : SUCCESS in 16m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aee5770865974e3c87f43fc9984b3b4b : SUCCESS in 5m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/093b8653e7664664910e8a8dc469b9e5 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/409621ff208544478f705268dbf1ce66 : SUCCESS in 6m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6fa89f46907049d094ac40f1fd98da84 : SUCCESS in 5m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1f23cd582ae54436892ef7dc23402722 : SUCCESS in 40m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3db6e41c345c4ee1b5b8fbbe7de009cb : SUCCESS in 6m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/620e9d76ee94420086380ac53f0ea2fe : SUCCESS in 1h 13m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/521b78410cad448c8111a9bd8f2bff14 : FAILURE in 1h 47m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/73b7dac99fbc4d78890131a5f4e82aae : SUCCESS in 1h 48m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/202d06e35b874343ab6698b5eea45a18 : SUCCESS in 2h 14m 22s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a0d67ee13c1147eea0c6bd75bc6c0917 : SUCCESS in 1h 05m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a55f844d0cf04d1690d34df17fc2cd5e : SUCCESS in 1h 27m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9c9bc68fac4545ac8a7ac1b909213bdf : SUCCESS in 44m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/15e61d4882804e28803f33bad055c86c : SUCCESS in 1h 15m 31s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2a93ce40d0fb4a379591df193d77de7b : SUCCESS in 1h 24m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e2dba63550ad4716a186c6fc7ca5412b : SUCCESS in 2h 26m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e31f7b119fa46e7b350a1cdc3228090 : SUCCESS in 46m 09s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe0be173dde84f009340088f0c1b5d1a : SUCCESS in 1h 12m 49s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/3fc25353bbee4623bee5b3f6a08715a7 : SUCCESS in 12m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1605877a4be09aa0d233e2dcb1a6f5384d5d9102","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-05-08 22:16:15.000000000","message":"Patch Set 10:\n\nBuild failed.","accounts_in_message":[],"_revision_number":10},{"id":"fcd766ceba282b048f5c415ab3182d5e66d6adda","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-05-08 22:31:36.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-847132-10/FC/ : SUCCESS in 2h 13m 50s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-847132-10/iSCSI/ : SUCCESS in 2h 44m 11s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"ea03cf6741dac8c4ffde0f6d8c04f0c86098f9fe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-08 23:34:04.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/847132/10/check/DellEMC_Unity_FC/17cd24d/DellEMC_Unity_FC/106 : SUCCESS in 1h 34m 45s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/10/check/DellEMC_Unity_iSCSI/439de32/DellEMC_Unity_iSCSI/109 : SUCCESS in 1h 55m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"cb47b404f5b3a3d642c788ee6f797c9846f79df1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-08 23:46:49.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-847132-10/ns5_nfs : SUCCESS in 3h 51m 40s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-847132-10/ns5_iscsi : SUCCESS in 3h 50m 47s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"8d306ea69d01dd58b191848c5140e9b326483db2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-09 00:06:29.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/10/upstream-check/cinder-tempest-plugin-ontap-fcp/955f71c/ : RETRY_LIMIT in 8m 31s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/10/upstream-check/cinder-tempest-plugin-ontap-iscsi/c539342/ : SUCCESS in 40m 57s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/10/upstream-check/cinder-tempest-plugin-ontap-nfs/570f162/ : SUCCESS in 46m 03s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/10/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/91ffaab/ : SUCCESS in 41m 36s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"3ee9b2639e6be36a4a8bf19dd141adb7bd067ee4","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-05-09 00:37:21.000000000","message":"Patch Set 10:\n\nBuild failed.","accounts_in_message":[],"_revision_number":10},{"id":"e569b3af3ddb6db606988c48d471326df648b11e","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-05-09 00:37:22.000000000","message":"Patch Set 10:\n\nBuild failed.","accounts_in_message":[],"_revision_number":10},{"id":"8237a16a9c753dc8d9b38f944aa1b59c082381e7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-05-09 04:16:41.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/32/847132/10/check/tempest-dsvm-full-focal-py3/0eb0e0d/ : FAILURE in 1h 34m 04s\n- devstack-plugin-ceph-dsvm-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/32/847132/10/check/devstack-plugin-ceph-dsvm-focal-py3/48582a9/ : FAILURE in 1h 02m 09s","accounts_in_message":[],"_revision_number":10},{"id":"d4a8cad68e042eaa6ce5c7575b33aefe59623e64","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-05-09 06:37:06.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/10/check/DellEMC_PowerStore_iSCSI/dc718a4/DellEMC_PowerStore_iSCSI/30 : SUCCESS in 52m 48s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/32/847132/10/check/DellEMC_PowerStore_FC/f88590e/DellEMC_PowerStore_FC/31 : SUCCESS in 1h 05m 42s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/32/847132/10/check/DellEMC_PowerStore_NFS/108129a/DellEMC_PowerStore_NFS/29 : SUCCESS in 54m 49s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/32/847132/10/check/DellEMC_PowerStore_NVMeTCP/5b4ebfc/DellEMC_PowerStore_NVMeTCP/29 : SUCCESS in 1h 08m 14s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"9d854c2fce16d53304b44f5c7e3db5c7f8476b06","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-05-09 09:33:12.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/32/847132/10/check/DellEMC_PowerFlex/4e4a312/DellEMC_PowerFlex/73 : FAILURE in 1h 35m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"241775c898becddcbd29770b5786aa00e11aabad","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-05-09 17:45:39.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/10/20230509235945/NEC-V-ISCSI : SUCCESS in 2h 40m 49s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/10/20230509235945/NEC-V-FC : SUCCESS in 2h 45m 51s","accounts_in_message":[],"_revision_number":10},{"id":"93119b7d839136530f7016200dcadf23477bca4b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-05-10 06:33:57.000000000","message":"Patch Set 10:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/10/20230510082935/NEC-ISCSI : SUCCESS in 3h 13m 11s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/10/20230510082935/NEC-FC : FAILURE in 7h 04m 19s","accounts_in_message":[],"_revision_number":10},{"id":"30c0e893947fe1ec4896a9e6a73428ec92510302","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-10 09:28:44.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"fa3c06dbe7e3103c22b9a28e78041eb56c8c83fd","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-05-13 12:03:07.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/32/847132/10/ : RETRY_LIMIT in 3m 05s","accounts_in_message":[],"_revision_number":10},{"id":"cca8f59c4de3a71a908a80b25b7e0f8d644881cb","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-05-19 14:23:08.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"34290053298a749208dcf7a32878df4afff8504f","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-05-19 15:15:12.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"fb509cb13d5d01ee13511af3eb9b490636901fe2","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-05-29 10:12:25.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e06c2053dfb73e61b15f5dc934fd221b0076369a","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-06-02 06:10:08.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"cb9e61f35f9716eacb54f293e43d81c073d0c705","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-06-02 06:17:55.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ad7b40c7c7da07aa049687cdb5512c4424102e91","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-07-21 13:25:08.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ea524dfc49983a7ad59b6b82fbbb8138a45fad7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-12-06 06:26:30.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"b58c409c4575b2f1edca296f737f06fbd1ce94b8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-12-06 06:26:43.000000000","message":"Patch Set 11:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16232/","accounts_in_message":[],"_revision_number":11},{"id":"b84b9d47c08bfa805445a79ecd51fb16b777dfb0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-12-06 06:26:44.000000000","message":"Patch Set 11:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16232/ : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"5ab9cd0dc2d2c6eaa2d088a643fd55792faaf5ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 06:53:32.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/914a4ba292154fbfb550cd31e4ced5b7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d533d1ac300741d5b31e4227b95ece99 : SUCCESS in 24m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6987bab645254dcd94337d2e0a4c206a : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e6f3c5cf2a294077990360e383b8c918 : SUCCESS in 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"87db8a4ac3d62d823e05bd582d0947fa6759bea1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-06 08:42:51.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2d1f79fc11b547b2988e2797cf2cf7f8 : SUCCESS in 1h 09m 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/aaa4dbbe8b0f46b08ccd68e66f354965 : SUCCESS in 1h 06m 43s","accounts_in_message":[],"_revision_number":11},{"id":"b63c7ff9d9be9996c1e6bea6f9b9779b9a88f475","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 08:59:18.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4ba7bc677684f408e73bacf024f698b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4996662bc4f24e61873233ad50d58167 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9141d1292ddf46e58c5bf25b1fe92287 : SUCCESS in 14m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a17ee272fc5444888dd8c8e99cee1ce8 : SUCCESS in 11m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faf8789c7f2b4afabd28477853c48ac7 : FAILURE in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/cdc4a5796c1445fd9a57d3c5e6a90c7c : SUCCESS in 50m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c01a0796ceed458a9ea5f8c7afe75cbb : SUCCESS in 56m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/62512f629e04412a946ca9b3326b368e : SUCCESS in 1h 17m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d7ae9decf8f4a57bf41edde4eaaace7 : SUCCESS in 1h 06m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77994bb6e6cb4ed181c7a464585ef4d9 : SUCCESS in 8m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ee51eb41c8bd4287a177c9a29a426c50 : SUCCESS in 15m 55s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2196e342a63d402390fc6f320af903c0 : SUCCESS in 12m 38s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/baf904d98e2e4561878310cd58913a9c : SUCCESS in 6m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5cae9d128e9b45178b45e5fe810dfec3 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ffd39c235de547b1a93b827e80763656 : SUCCESS in 6m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d0f3cfc3ee043c09b896eb5af9414c0 : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/421521a5346e44aeb3d5d8477cea8fc7 : SUCCESS in 48m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0dfd052494f44fd3a4caaa15b7bbf0b7 : SUCCESS in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/60ee6aec6cef4b7ab9a57e12f9c7b76a : SUCCESS in 2h 20m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7ff5ef613692426c94875469dd8ae589 : SUCCESS in 2h 02m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1cf4bf3629ac4efdb8c35b423183274b : SUCCESS in 1h 46m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/56464258c21548a29e7a9d652c2f02a6 : SUCCESS in 2h 08m 04s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ce6ff985f63845e8a721212b39ce5436 : SUCCESS in 1h 18m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f9561c0bc8804e18ae0eab791a595f4b : SUCCESS in 1h 08m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/edc7600c249f4b6f8c9ef9db60343865 : SUCCESS in 47m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f9a443dc4769461c9ae19466fd01ad86 : SUCCESS in 1h 22m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/186d975d87b1496b80bcdb2fc497ffe2 : SUCCESS in 1h 26m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0a414ebc9a2c49eeb140736379055105 : SUCCESS in 1h 23m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5a736a36db74de2884e624844832137 : SUCCESS in 1h 03m 59s","accounts_in_message":[],"_revision_number":11},{"id":"3d9d8e7c9b573fb2f7c26c8e10f81b7bc9fa203d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-12-06 09:00:14.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/633d31a93439445da62cab690664c929 : RETRY_LIMIT in 1m 04s\n- pure-devstack-cinder-tempest-fc-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/25a9c2e6e7f541439f95d29b62604265 : RETRY_LIMIT in 1m 03s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/6498c9b20043414badc0401be03f3892 : RETRY_LIMIT in 1m 03s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/dcdb082f636c4a7a8f86c24e4d7d60f1 : RETRY_LIMIT in 1m 05s","accounts_in_message":[],"_revision_number":11},{"id":"6c238b6522becdefda754ebc92fe2953e9ff138a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-12-06 12:38:27.000000000","message":"Patch Set 11:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-847132-11/FC/ : FAILURE in 2h 11m 31s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-847132-11/iSCSI/ : UNSTABLE in 3m 16s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":11},{"id":"7dac076a0434683820300ea8d21099fb9823813b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-12-06 13:50:42.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/c6fc608908bc4c2bbe1a35801a391c39\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/402a60fe6ff0418d84bfead08aaad9f9 : FAILURE in 1h 24m 43s","accounts_in_message":[],"_revision_number":11},{"id":"783e76396ab807ae4626fffb83bc5345b1743a39","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-12-06 14:32:22.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"2969aa4dd4c1c9ab094b17a66619364aaac937dd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-06 17:35:28.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/32/847132/11/check/DellEMC_PowerFlex/222960e/DellEMC_PowerFlex/1000 : SUCCESS in 1h 26m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"2816f8605d93d206eb4be7e4086d5d539b10c4c5","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-12-07 00:18:09.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/32/847132/11/ : SUCCESS in 1h 14m 47s","accounts_in_message":[],"_revision_number":11},{"id":"0c2b4f8264ccaeb54d13bfe603f200628b4547d8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-07 01:09:32.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/633e70067aaf4b97a2515f69c55b8b1e : SUCCESS in 1h 08m 44s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/96af11d800bf4b588e18b0ae749c9fa5 : SUCCESS in 1h 04m 34s","accounts_in_message":[],"_revision_number":11},{"id":"4a3c569c018bb3687b223aa88656a2f8144684d3","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-12-07 12:02:32.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/11/20231207172257/NEC-V-ISCSI : SUCCESS in 3h 17m 47s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/11/20231207172257/NEC-V-FC : SUCCESS in 3h 39m 33s","accounts_in_message":[],"_revision_number":11},{"id":"8ed22e832330d93ba7873225e3c0548b2e18b924","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-07 17:47:35.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/847132/11/check/DellEMC_Unity_FC/b2c5500/DellEMC_Unity_FC/1054 : FAILURE in 2h 55m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/11/check/DellEMC_Unity_iSCSI/aab19c9/DellEMC_Unity_iSCSI/1068 : ABORTED in 6h 00m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"8b88ea4d79d7ff4e472419b399082a9de0302dd3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-12-08 06:43:58.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/11/20231208053457/NEC-ISCSI : SUCCESS in 4h 02m 17s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/11/20231208053457/NEC-FC : SUCCESS in 10h 08m 57s","accounts_in_message":[],"_revision_number":11},{"id":"1930cecc2f803c872f1a7784459233470cf358cb","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-12-11 07:28:38.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5c47996c6efb1572444b7a7ab99673ae552891b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2024-01-18 12:56:47.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"10acce1d64dad467519a128800eacae83b7dd850","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 13:39:00.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e35f13c7e94349609d0763500c9f46f4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e0e6f24d9b24b678f9a95926f4ce51b : SUCCESS in 14m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/41182292e1084a8dbbe68407b1a468cf : FAILURE in 38m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/922724c591c54a19b17fb4b216ff99be : SUCCESS in 11m 21s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"11741423dd294ae560e04a89ee47eda96ac01699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 15:14:16.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac77a178c5f24920ac1a22339f7634b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aa65c6950a7428b88a3c231001d8f68 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/721e1b8cbe7749678592796b7b13455c : SUCCESS in 10m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b03ff335066463a8ae3b5efee32e12d : SUCCESS in 10m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47995581285e40988877a03d3de6cf8d : FAILURE in 6m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/db97028ffd3e461d92b87d1f29916f09 : SUCCESS in 1h 04m 01s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4e175426329147c3ae2b38805d262670 : SUCCESS in 51m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ee4c153dcea74ffaa7c28f50e982a8d9 : SUCCESS in 1h 20m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ed834d177714e2a93c6294c90a338ea : SUCCESS in 26m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21e001d763674f8bae7034f81b5f45c4 : SUCCESS in 9m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6f0cef0428df4aab9830e3cd105f0c60 : SUCCESS in 17m 23s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a479930cd0c64324bdf88b6474f19d1f : SUCCESS in 12m 40s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aed92755a44c421c91327065c4794ab6 : SUCCESS in 5m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/92aab68c257744e783635e09d0039306 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c0837acc394f4080853770c6855c894b : SUCCESS in 6m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ad4d5de47c4c40d28950641f7dccf2c2 : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a7dde3bde7f14e3889c28863ee1c2d28 : SUCCESS in 46m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/55d232da8f104677a3bdff5db1f07536 : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/efb5151c68c140b9838fbb389d918bbd : SUCCESS in 1h 42m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/405345a2f2704527b1e35228767ed778 : FAILURE in 1h 55m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/25dd89a677f7463fa4bce26b31b3820a : SUCCESS in 1h 49m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6477a4d4010442f1b9d880c016f463be : SUCCESS in 2h 10m 47s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f81bbd8e41b9404b8fb4599609c8c429 : SUCCESS in 1h 21m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/76889576abc645ffb54c4bf225cbfb1a : SUCCESS in 1h 41m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c52f71ad2aaa4c6faba4cc347effdb49 : SUCCESS in 37m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/67edead6d87648ca85ac375a415860ea : SUCCESS in 1h 32m 56s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4d4ae23ed9734efcb6d1add2747b0a3d : SUCCESS in 1h 13m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/73ebd08f0fa74b249e400b1884f0bb2e : SUCCESS in 1h 50m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/970a36bc0af14fa189c2e1f3db42c664 : SUCCESS in 53m 08s","accounts_in_message":[],"_revision_number":12},{"id":"f92da0fb9eb3b07ef49a4bde0f0e0ae16cd422a9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-01-18 19:11:57.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1ae8dd52fef64dd7898861ed464239d0\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0e5d9a6526ea42b5bd3dfffe26f9c209 : FAILURE in 1h 26m 46s","accounts_in_message":[],"_revision_number":12},{"id":"6e6e47d33b98f568559e113ea703e91f656202bd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-01-18 22:25:46.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847132 : FAILURE in 50m 06s","accounts_in_message":[],"_revision_number":12},{"id":"90b94b68ba1a53004d9a61d5fbdf5a50e51f47db","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-01-18 23:05:25.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bdde2ade3b254b4c8102e823febafa2b : SUCCESS in 1h 07m 56s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6a893c7d33594bf8a67f6d0c193662a8 : SUCCESS in 1h 17m 45s","accounts_in_message":[],"_revision_number":12},{"id":"ecc933582cefe52dc3c9a795be899c03a407b54f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-19 06:27:24.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/466/.","accounts_in_message":[],"_revision_number":12},{"id":"fc4eddd13e60ac0b70272c301de984e82a664a4e","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2024-01-19 14:18:55.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"bbe96d0df3af0538e5ad3486f9ddccd5afb3ccd1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-19 16:15:39.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/32/847132/12/check/DellEMC_PowerFlex_3.6/6d92381/DellEMC_PowerFlex_3.6/1253 : SUCCESS in 1h 14m 41s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/32/847132/12/check/DellEMC_PowerFlex_4.5/d01865b/DellEMC_PowerFlex_4.5/224 : FAILURE in 2h 09m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"6bb7d8ce7b541bee03b6febf0062aec8de193c09","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-19 16:54:27.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/463/.","accounts_in_message":[],"_revision_number":12},{"id":"516ad0bf228153abe46b861b759ceecd7ef7f4e5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-01-22 04:48:48.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/12/20240122073642/NEC-V-FC : SUCCESS in 3h 47m 19s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/12/20240122073642/NEC-V-ISCSI : SUCCESS in 6h 12m 02s","accounts_in_message":[],"_revision_number":12},{"id":"a7cbbb9fcea077bc3af89e0e3be54fed5c8e5b36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-25 07:33:12.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"33903db9df0f66b303ccf596b73d990e1e031d3c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-25 07:33:23.000000000","message":"Patch Set 13:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/142/","accounts_in_message":[],"_revision_number":13},{"id":"00e1c109e5386af984974fae1b6a7c0454ab8f45","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-25 07:39:38.000000000","message":"Patch Set 13:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/142/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":13},{"id":"91807a9ec011275d258b773f695b7c818811cccb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 07:43:01.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a188bd0bcc7043b5907a680152328dde\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb8d9bcdecf2488bae5103f966da9bb3 : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c343605d16944426bda9f06dd15d1a54 : SUCCESS in 7m 56s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8b619dde30f4543cc768b5a920f78596d4370e10","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-25 08:08:49.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/847132 : FAILURE in 33m 31s","accounts_in_message":[],"_revision_number":13},{"id":"e0779e1cac536e6ce8558c5360bd87b0fe1777c1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-25 08:34:00.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/847132 : SUCCESS in 41m 19s","accounts_in_message":[],"_revision_number":13},{"id":"96f28fb1406c64d629078639c5512ff3c7a3c44e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-25 08:36:09.000000000","message":"Patch Set 13:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/77bd3b5cc6b14b9c9ed0f66b563ff51e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/d4b42e044f5e4584a270a7c3175cf40d : POST_FAILURE in 21m 27s","accounts_in_message":[],"_revision_number":13},{"id":"5385fde1182cdc8b20410133a8359f46279241a7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-25 08:54:34.000000000","message":"Patch Set 13:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/788/.","accounts_in_message":[],"_revision_number":13},{"id":"a4861d437a619ab092f63a18b18998796886d8e5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-25 08:55:22.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/77e3e393e0aa480e87177a1904777326 : FAILURE in 1h 20m 13s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/eec320aac0054b5ab8f9b64e930af9c0 : SUCCESS in 1h 18m 16s","accounts_in_message":[],"_revision_number":13},{"id":"6bb9ceb872daa6ba98a9a3aef338bd99021b6883","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-25 09:52:48.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/847132/13/check/DellEMC_Unity_FC/2d0df3c/DellEMC_Unity_FC/2294 : SUCCESS in 1h 32m 58s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/13/check/DellEMC_Unity_iSCSI/944482e/DellEMC_Unity_iSCSI/2301 : SUCCESS in 2h 03m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"0514f86e4c45a4b8d34fb79d880c5582df79935f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 10:30:09.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c71728cf30a4977948137c736b7808a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/352367de57ef4422a830ab43618b6cbf : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b145069758374fde84bbeb288e07842d : SUCCESS in 11m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9d689d5ee3054175b310928c35a724a8 : FAILURE in 10m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5492f54f4dff4893b8d31102fc7d0c83 : SUCCESS in 15m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cd2b93a0a884fa2984312c2953b1e77 : SUCCESS in 12m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/a9fea53675ce41a2afaa9db25ba15cac : SUCCESS in 57m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e4b8a3c170f2439f9744a329fda603bc : SUCCESS in 1h 21m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80e37df4507c4726924bd0d77337b5d6 : SUCCESS in 59m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7e32309fba943d0bc622d09e7b6cc7f : SUCCESS in 17m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/391ae49ccc1b4458922a700f899a5124 : SUCCESS in 16m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cfc89124125d404b878df00b9aa569dc : SUCCESS in 5m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bac85388e3fc450d9c21529e5bf04686 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/83ae94c9ddda4b73a07f7a3625ebca66 : SUCCESS in 6m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1241800faa054111836aef1f585ca7e4 : SUCCESS in 6m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/be79fac3cc55405789e2f88477e73c5d : FAILURE in 42m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ed391c8916744fe3ac2643f94b842255 : SUCCESS in 5m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4610d075295e4797bbb9115d557c346e : SUCCESS in 1h 57m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/61efb6247771414f9b755f576baf6a71 : FAILURE in 2h 50m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bb364f719424453eb7ddbe2002733014 : SUCCESS in 2h 01m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/08317095bc6d4a9d827a7db64e5d31c0 : SUCCESS in 1h 42m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/eaa94aa39cea45049cf1eff393aa83e2 : SUCCESS in 29m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/07422713ed1e4a508a6b13cc8ae93013 : SUCCESS in 1h 16m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f664c3c80a81488dac32cea843ab0833 : SUCCESS in 1h 43m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/217856de251b48cf8550d28dec804acf : SUCCESS in 40m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a023f71c48ee480ca9be9693f6ceb6a8 : SUCCESS in 1h 18m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a4961583b7df4318b111a805bc8366bd : SUCCESS in 1h 12m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d4b08390fc9049ba9d939752a787a079 : SUCCESS in 1h 21m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3af5397e12b74b06a66c5eb5ab2fc965 : SUCCESS in 49m 56s","accounts_in_message":[],"_revision_number":13},{"id":"a7f8821d0fcef6025fdb0df2ab0692038b94dc5b","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-25 10:47:58.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/847132 : TIMED_OUT in 3h 00m 36s","accounts_in_message":[],"_revision_number":13},{"id":"bd1c7945341252e07c16968e345b3228d5ac8d88","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-25 10:50:52.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"8471af78a43cf568ad146fb97553634af940012e","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-25 11:27:13.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"2195b1c5d217930810ff697c34fe3396320d658c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 11:50:34.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d6b49847b9c4d04b78e302e90554e8a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7fbf5fd8887c4c668d0201c6249e2905 : FAILURE in 21m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/05a59de9b8f04c8d93c0eae54332729c : SUCCESS in 18m 18s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8310462b8afe6b7383948be838cd0bfd5a40d240","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-25 11:52:45.000000000","message":"Patch Set 13:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/6106aae79a0a4370b5da68d7afe8c5c1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e31772441c7a4d0aaa06f40c8c936323 : POST_FAILURE in 20m 58s","accounts_in_message":[],"_revision_number":13},{"id":"0541dbd19206c4807d2eed9a04ee9d9dcb6820d8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-25 12:43:52.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3f9e8c31fb96439fa41c78dae8b4d539 : SUCCESS in 1h 14m 36s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f9c65b8810784a05bb95c79bf7ef034b : SUCCESS in 1h 11m 46s","accounts_in_message":[],"_revision_number":13},{"id":"988390ff6f8f8c531f4ffd2b29918834ccda0800","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-25 12:56:44.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/32/847132/13/check/DellEMC_PowerFlex_3.6/2f2c467/DellEMC_PowerFlex_3.6/1782 : FAILURE in 2h 53m 55s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/32/847132/13/check/DellEMC_PowerFlex_4.5/c864174/DellEMC_PowerFlex_4.5/661 : SUCCESS in 1h 42m 43s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/32/847132/13/check/DellEMC_PowerFlex_4.x_NVMeTCP/bf00c4e/DellEMC_PowerFlex_4.x_NVMeTCP/619 : FAILURE in 14m 37s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"bc2f218911f9a23afab92ad4e3f39a6a06157209","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-07-25 12:58:06.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"2f005ac0250c9c575f9e787f5f97ee049a5b9dd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 13:48:05.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba238aa61dee4fa3a676101ed1ecd7c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a44bd5bb167e447c96118e4761d2406c : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ade4e0a857d944ac939d5fa5e4a99a08 : SUCCESS in 11m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/715c548373dd422fa1e33c1f47e91c33 : SUCCESS in 11m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ce305c3076740b6aff039f439bfc92e : SUCCESS in 17m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/175bce13c7294744b2fa3a4d9e0b99f7 : SUCCESS in 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/e5297e6702044e5087c02acc949cdc89 : SUCCESS in 56m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e122eab99a3248cbac47a1f36c99a780 : SUCCESS in 1h 34m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48b28d5a6e3147cb895cbfab866d1979 : SUCCESS in 57m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4346b31a3e1457dbc6b946d06706ccc : SUCCESS in 19m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e543f3d2d8884077ad12aeb3de00e4f7 : SUCCESS in 16m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/60fd5bf0a0c9485daf567208f8158fe8 : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/df119a2ece324c7f8273e8fa6bcd2fe3 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2783e3da46f641ad9e766185fc17dae6 : SUCCESS in 5m 32s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3c658d9fc4484804aa3117939e15cd80 : SUCCESS in 5m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5dcf3ef3375b42cd9b007b84003a9176 : FAILURE in 44m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7f3b19503f0141b8b4596e224580bf61 : SUCCESS in 5m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/01ae4173cd9946629dacd8cdaf9261ad : SUCCESS in 2h 01m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0f810e9516634714b8590b95d0535125 : FAILURE in 2h 12m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cb3aff8e1598475292b692fa596bd5e7 : SUCCESS in 1h 58m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/628f1de174c94132885b68c66d0677b8 : SUCCESS in 2h 03m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/15c91c62fbbf44ed836998e2066b41f5 : SUCCESS in 33m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ddcdb102cd4d46c3bb8daeb66dfb2161 : SUCCESS in 1h 25m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7be1209f5cca45d282e1c6216c732ace : SUCCESS in 1h 29m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b100a6a389b14470b1d93e6f66b1693f : SUCCESS in 43m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/655b7fba332f4953a8a188dbd1c0d434 : SUCCESS in 1h 32m 47s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/bc6ffa2511564e49945b449a35b081a7 : POST_FAILURE in 2h 15m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/03a09898344c4fef846c9c84d79b9b04 : SUCCESS in 1h 31m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c61ba30915e147ee9e8837ba7f471984 : SUCCESS in 49m 21s","accounts_in_message":[],"_revision_number":13},{"id":"0663db3c22f5bde4a04d982ac150b6873d2d4525","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-25 13:53:45.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/25_07_2024_13_48_16\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":13},{"id":"0de9df2a80d63bb151d9f8ceb88f0d393a101dd3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-25 13:57:17.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/13/upstream-check/cinder-tempest-plugin-ontap-iscsi/8f823bc/ : RETRY_LIMIT in 8m 32s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/847132/13/upstream-check/cinder-tempest-plugin-ontap-nfs/704d3f6/ : RETRY_LIMIT in 8m 00s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"9391288e8d3cca9125b04488bc9e45ac6b221189","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-25 15:47:05.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/13/check/DellEMC_PowerStore_iSCSI/ba0d423/DellEMC_PowerStore_iSCSI/1542 : SUCCESS in 1h 17m 05s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/32/847132/13/check/DellEMC_PowerStore_FC/c78c95b/DellEMC_PowerStore_FC/1738 : SUCCESS in 1h 02m 12s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/32/847132/13/check/DellEMC_PowerStore_NFS/42c171f/DellEMC_PowerStore_NFS/1236 : FAILURE in 1h 17m 22s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/32/847132/13/check/DellEMC_PowerStore_NVMeTCP/a01b3fe/DellEMC_PowerStore_NVMeTCP/1095 : SUCCESS in 1h 58m 41s\n- DellEMC_PowerStore_NVMeFC https://elab-os-logsrv.delllabs.net/32/847132/13/check/DellEMC_PowerStore_NVMeFC/bc5e113/DellEMC_PowerStore_NVMeFC/59 : SUCCESS in 1h 07m 28s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":13},{"id":"cbdc638611b5204345c0ee91e55982de9e652633","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-25 16:26:59.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/13/20240725213807/NEC-V-FC : SUCCESS in 3h 46m 45s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/13/20240725213807/NEC-V-ISCSI : SUCCESS in 3h 48m 49s","accounts_in_message":[],"_revision_number":13},{"id":"5cb7252882a39d12b2657a6ebdfee2e19f36544f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-25 17:40:34.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/847132/13/check/DellEMC_Unity_FC/c46a3e3/DellEMC_Unity_FC/2298 : SUCCESS in 1h 38m 11s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/847132/13/check/DellEMC_Unity_iSCSI/ca203ea/DellEMC_Unity_iSCSI/2305 : SUCCESS in 2h 10m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"55c7a59f873a6813ff880d80c335eb43bde6fadc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-25 21:42:08.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/32/847132/13/check/DellEMC_PowerFlex_3.6/f610e1f/DellEMC_PowerFlex_3.6/1786 : FAILURE in 2h 55m 31s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/32/847132/13/check/DellEMC_PowerFlex_4.5/a176037/DellEMC_PowerFlex_4.5/665 : SUCCESS in 2h 06m 09s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/32/847132/13/check/DellEMC_PowerFlex_4.x_NVMeTCP/001bc11/DellEMC_PowerFlex_4.x_NVMeTCP/623 : FAILURE in 13m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"2b5f7fbe71b893a00a5a0ab79eacfb9e79b6941b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-26 11:57:45.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/13/20240726120814/NEC-ISCSI : SUCCESS in 5h 15m 10s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/847132/13/20240726120814/NEC-FC : SUCCESS in 8h 49m 27s","accounts_in_message":[],"_revision_number":13},{"id":"6d63923f353e365bcfc02d7663b8289c21207301","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-07-29 21:30:01.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"e5b6641041e5535c2a560b2ecbfc0de241cc65fd","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-07-31 14:32:00.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"b347edc9dc3a501bc8eb566266af1bf0b51a609d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 14:32:47.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"1c3783e9c3ca330b0384c75299e4f2356f72a991","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 16:49:03.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55b22240e42646ca99af3a8f255cc15a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee836194d31840c29fd99d864e641f59 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fcf486acaca4449bc106d814c6b31d3 : SUCCESS in 10m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85618db0733c4d0da7677f0bd1aa2b28 : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/982e9d9d0da64f7d929f192537297220 : SUCCESS in 11m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/3f2eca2b29fe4fe0ab88cb27c9d76076 : SUCCESS in 54m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd9b9830b06c4f228270dbffda16afc3 : SUCCESS in 1h 21m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74ac932aa2934689a687f2754d6b95c5 : SUCCESS in 1h 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f995618739a24619bcfb017be222b262 : SUCCESS in 20m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/031c83ade3714f2581e59ea366db607e : SUCCESS in 1h 11m 41s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/190d0ee249ed4ec3a5bb141deb4957a0 : SUCCESS in 1h 54m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44949d40b99c4685b72373331855eb4f : SUCCESS in 49m 42s","accounts_in_message":[],"_revision_number":13},{"id":"c57c4f55e59a73ef493678ca4b0c230d562f5848","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 16:49:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"f3140c69957cbe7044f95731431c7a741906da7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 16:51:11.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/075ba9daa58548cabb0accffab34ce7d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b8cece3c7794ef5a87d57758c893f79 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6aa98c95af194fa28cf4ff60e873a46b : SUCCESS in 42s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"c0e31e787f3d5f76aa300972f656fd22fdc76601","revisions":{"fbda4dbe0f2c495a821cc10dfedbb0b171ac50bc":{"kind":"REWORK","_number":1,"created":"2022-06-22 07:58:43.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/32/847132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/847132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/847132/1"}}},"commit":{"parents":[{"commit":"78a748a4abd7aa1aa56256222754a1206001af12","subject":"StorPool: create_cloned_volume() improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78a748a4abd7aa1aa56256222754a1206001af12"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:48:25.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:57:23.000000000","tz":180},"subject":"StorPool: declare the clone_across_pools capability","message":"StorPool: declare the clone_across_pools capability\n\nChange-Id: I5338c6c4f53a448e495f695cd64b36b722cd947d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fbda4dbe0f2c495a821cc10dfedbb0b171ac50bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fbda4dbe0f2c495a821cc10dfedbb0b171ac50bc"}]},"branch":"refs/heads/master"},"f644bd552d6b7d0ab42b23aade80db67d0f4a4a3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-06-24 07:50:17.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/32/847132/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/847132/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/847132/2"}}},"commit":{"parents":[{"commit":"30dd73246a5e6b023b9b389ebf5b0b80b1f0b15a","subject":"StorPool: create_cloned_volume() improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/30dd73246a5e6b023b9b389ebf5b0b80b1f0b15a"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:48:25.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-24 07:50:01.000000000","tz":180},"subject":"StorPool: declare the clone_across_pools capability","message":"StorPool: declare the clone_across_pools capability\n\nChange-Id: I5338c6c4f53a448e495f695cd64b36b722cd947d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f644bd552d6b7d0ab42b23aade80db67d0f4a4a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f644bd552d6b7d0ab42b23aade80db67d0f4a4a3"}]},"branch":"refs/heads/master"},"547225b7d043ac334602e1a6b9ed74faba84e542":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-08-01 13:13:55.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/32/847132/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/847132/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/847132/3"}}},"commit":{"parents":[{"commit":"dec5acd4ff4c13a513372dc8ee48eafae0e15e77","subject":"StorPool: create_cloned_volume() improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dec5acd4ff4c13a513372dc8ee48eafae0e15e77"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:48:25.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-08-01 13:13:43.000000000","tz":180},"subject":"StorPool: declare the clone_across_pools capability","message":"StorPool: declare the clone_across_pools capability\n\nChange-Id: I5338c6c4f53a448e495f695cd64b36b722cd947d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/547225b7d043ac334602e1a6b9ed74faba84e542"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/547225b7d043ac334602e1a6b9ed74faba84e542"}]},"branch":"refs/heads/master"},"71e6a2faa5ad0feb972dbbade4b81a6cb095fb40":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-08-02 08:04:22.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/32/847132/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/847132/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/847132/4"}}},"commit":{"parents":[{"commit":"a2285f2d55bb97ae1e203fe240043b013e8b46da","subject":"StorPool: create_cloned_volume() improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2285f2d55bb97ae1e203fe240043b013e8b46da"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:48:25.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-08-02 08:04:08.000000000","tz":180},"subject":"StorPool: declare the clone_across_pools capability","message":"StorPool: declare the clone_across_pools capability\n\nChange-Id: I5338c6c4f53a448e495f695cd64b36b722cd947d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/71e6a2faa5ad0feb972dbbade4b81a6cb095fb40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/71e6a2faa5ad0feb972dbbade4b81a6cb095fb40"}]},"branch":"refs/heads/master"},"c70efd1652d8ad5874f7c704d192a08d31f32602":{"kind":"REWORK","_number":5,"created":"2023-01-16 14:56:42.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/32/847132/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/847132/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/847132/5"}}},"commit":{"parents":[{"commit":"b10b10f930edbc66e24c82012ede697f379e4241","subject":"Add the clone_across_pools driver capability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b10b10f930edbc66e24c82012ede697f379e4241"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:48:25.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-01-16 14:34:22.000000000","tz":120},"subject":"StorPool: declare the clone_across_pools capability","message":"StorPool: declare the clone_across_pools capability\n\nChange-Id: I5338c6c4f53a448e495f695cd64b36b722cd947d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c70efd1652d8ad5874f7c704d192a08d31f32602"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c70efd1652d8ad5874f7c704d192a08d31f32602"}]},"branch":"refs/heads/master"},"c0d034b56f65113c65e563a4ac987c9d81e53f91":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-01-26 12:35:12.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/32/847132/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/847132/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/847132/6"}}},"commit":{"parents":[{"commit":"bbfffa38d7c87d4b83c42bcc92f50dc9fd028d73","subject":"Add the clone_across_pools driver capability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bbfffa38d7c87d4b83c42bcc92f50dc9fd028d73"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:48:25.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-01-26 12:33:35.000000000","tz":120},"subject":"StorPool: declare the clone_across_pools capability","message":"StorPool: declare the clone_across_pools capability\n\nChange-Id: I5338c6c4f53a448e495f695cd64b36b722cd947d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0d034b56f65113c65e563a4ac987c9d81e53f91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0d034b56f65113c65e563a4ac987c9d81e53f91"}]},"branch":"refs/heads/master"},"b257f9ca8c349d9d42d35e93350bb30f986d5e28":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-02-08 13:33:35.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/32/847132/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/847132/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/847132/7"}}},"commit":{"parents":[{"commit":"b09cc1fd5e5f9dbdbefb6966d4228e878a2fec12","subject":"Add the clone_across_pools driver capability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b09cc1fd5e5f9dbdbefb6966d4228e878a2fec12"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:48:25.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-02-08 13:33:11.000000000","tz":120},"subject":"StorPool: declare the clone_across_pools capability","message":"StorPool: declare the clone_across_pools capability\n\nChange-Id: I5338c6c4f53a448e495f695cd64b36b722cd947d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b257f9ca8c349d9d42d35e93350bb30f986d5e28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b257f9ca8c349d9d42d35e93350bb30f986d5e28"}]},"branch":"refs/heads/master"},"322c67aa22d79e34110c81d9c673eaee23258aaf":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-02-13 09:20:32.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/32/847132/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/847132/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/847132/8"}}},"commit":{"parents":[{"commit":"c6e23b7eb83d0b0a852d46f0fe87669fe9c0ea81","subject":"Add the clone_across_pools driver capability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6e23b7eb83d0b0a852d46f0fe87669fe9c0ea81"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:48:25.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-02-13 09:10:03.000000000","tz":120},"subject":"StorPool: declare the clone_across_pools capability","message":"StorPool: declare the clone_across_pools capability\n\nChange-Id: I5338c6c4f53a448e495f695cd64b36b722cd947d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/322c67aa22d79e34110c81d9c673eaee23258aaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/322c67aa22d79e34110c81d9c673eaee23258aaf"}]},"branch":"refs/heads/master"},"2b286cefd5def16663bce4f2cef4a38e3e93425a":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-03-01 12:49:08.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/32/847132/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/847132/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/847132/9"}}},"commit":{"parents":[{"commit":"c47c8054fc1f6239f050a0865d8ac8f8da1b073d","subject":"Add the clone_across_pools driver capability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c47c8054fc1f6239f050a0865d8ac8f8da1b073d"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:48:25.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-03-01 12:49:01.000000000","tz":120},"subject":"StorPool: declare the clone_across_pools capability","message":"StorPool: declare the clone_across_pools capability\n\nChange-Id: I5338c6c4f53a448e495f695cd64b36b722cd947d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b286cefd5def16663bce4f2cef4a38e3e93425a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b286cefd5def16663bce4f2cef4a38e3e93425a"}]},"branch":"refs/heads/master"},"26e09091de4e06485dbf0319d75e321450045980":{"kind":"REWORK","_number":10,"created":"2023-05-08 19:32:02.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/32/847132/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/847132/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/847132/10"}}},"commit":{"parents":[{"commit":"eeb95e0adfda075e395a15c1825448a92824008d","subject":"Add the clone_across_pools driver capability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eeb95e0adfda075e395a15c1825448a92824008d"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:48:25.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-05-08 19:21:35.000000000","tz":180},"subject":"StorPool: declare the clone_across_pools capability","message":"StorPool: declare the clone_across_pools capability\n\nChange-Id: I5338c6c4f53a448e495f695cd64b36b722cd947d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26e09091de4e06485dbf0319d75e321450045980"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/26e09091de4e06485dbf0319d75e321450045980"}]},"branch":"refs/heads/master"},"381f77449f30d9a860eeef532963e21f895c00e9":{"kind":"REWORK","_number":11,"created":"2023-12-06 06:26:30.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/32/847132/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/847132/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/847132/11"}}},"commit":{"parents":[{"commit":"8ce9bb75fb32c921043b14b43884c8b1660cd9b8","subject":"Add the clone_across_pools driver capability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ce9bb75fb32c921043b14b43884c8b1660cd9b8"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:48:25.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-12-06 06:22:54.000000000","tz":120},"subject":"StorPool: declare the clone_across_pools capability","message":"StorPool: declare the clone_across_pools capability\n\nChange-Id: I5338c6c4f53a448e495f695cd64b36b722cd947d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/381f77449f30d9a860eeef532963e21f895c00e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/381f77449f30d9a860eeef532963e21f895c00e9"}]},"branch":"refs/heads/master"},"d014301dd2f5bf06f2deca33c2e97bbdd4a99e24":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2024-01-18 12:56:47.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/32/847132/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/847132/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/847132/12"}}},"commit":{"parents":[{"commit":"c227eb238e464e31f3a08a1bdb2bab2bd844d50e","subject":"Add the clone_across_pools driver capability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c227eb238e464e31f3a08a1bdb2bab2bd844d50e"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:48:25.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2024-01-18 09:26:10.000000000","tz":120},"subject":"StorPool: declare the clone_across_pools capability","message":"StorPool: declare the clone_across_pools capability\n\nChange-Id: I5338c6c4f53a448e495f695cd64b36b722cd947d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d014301dd2f5bf06f2deca33c2e97bbdd4a99e24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d014301dd2f5bf06f2deca33c2e97bbdd4a99e24"}]},"branch":"refs/heads/master"},"c0e31e787f3d5f76aa300972f656fd22fdc76601":{"kind":"REWORK","_number":13,"created":"2024-07-25 07:33:12.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/32/847132/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/847132/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/847132/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/847132/13"}}},"commit":{"parents":[{"commit":"e73954d56dfe923d75a830ac37de2ec09424adbb","subject":"Merge \"Add the clone_across_pools driver capability\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e73954d56dfe923d75a830ac37de2ec09424adbb"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-06-22 07:48:25.000000000","tz":180},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-07-25 06:55:42.000000000","tz":330},"subject":"StorPool: declare the clone_across_pools capability","message":"StorPool: declare the clone_across_pools capability\n\nDepends-on: blueprint clone-across-pools\nChange-Id: I5338c6c4f53a448e495f695cd64b36b722cd947d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0e31e787f3d5f76aa300972f656fd22fdc76601"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0e31e787f3d5f76aa300972f656fd22fdc76601"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
