)]}'
{"id":"openstack%2Fcinder~848900","triplet_id":"openstack%2Fcinder~master~I7eda4645575cfaedcf45d73ab3a215976d3fac3a","project":"openstack/cinder","branch":"master","topic":"powrmax-deadlocks","attention_set":{},"removed_from_attention_set":{"33473":{"account":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"last_update":"2023-07-03 10:27:35.000000000","reason":"\u003cGERRIT_ACCOUNT_33473\u003e replied on the change","reason_account":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"}},"31779":{"account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"last_update":"2023-12-15 23:04:10.000000000","reason":"Change was submitted"},"35075":{"account":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"last_update":"2023-12-15 23:04:10.000000000","reason":"Change was submitted"},"34988":{"account":{"_account_id":34988,"name":"Andrey Pakhomov","email":"a.pakhomov@dell.com"},"last_update":"2023-12-15 23:04:10.000000000","reason":"Change was submitted"},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2023-12-15 23:04:10.000000000","reason":"Change was submitted"},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2023-12-15 23:04:10.000000000","reason":"Change was submitted"},"32238":{"account":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"last_update":"2023-12-15 23:04:10.000000000","reason":"Change was submitted"},"34290":{"account":{"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},"last_update":"2022-07-12 11:26:26.000000000","reason":"\u003cGERRIT_ACCOUNT_34290\u003e replied on the change","reason_account":{"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"}},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2023-12-15 23:04:10.000000000","reason":"Change was submitted"},"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2023-12-15 23:04:10.000000000","reason":"Change was submitted"},"33502":{"account":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"last_update":"2022-07-25 12:58:57.000000000","reason":"\u003cGERRIT_ACCOUNT_33502\u003e replied on the change","reason_account":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-12-15 23:04:10.000000000","reason":"Change was submitted"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2023-12-15 23:04:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7eda4645575cfaedcf45d73ab3a215976d3fac3a","subject":"PowerMax: Fix deadlock moving SGs","status":"MERGED","created":"2022-07-06 19:31:08.000000000","updated":"2023-12-15 23:05:40.000000000","submitted":"2023-12-15 23:04:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848900-powrmax-deadlocks","meta_rev_id":"c54b4b45ebf46f69d7a2732d4742804777e30fc0","_number":848900,"virtual_id_number":848900,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-15 23:04:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"all":[{"value":1,"date":"2023-07-11 19:56:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},{"value":2,"date":"2023-12-15 20:55:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2023-09-12 13:26:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},{"value":1,"date":"2023-12-15 20:55:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34988,"name":"Andrey Pakhomov","email":"a.pakhomov@dell.com"}],"reviewers":{"CC":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},{"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},{"_account_id":34988,"name":"Andrey Pakhomov","email":"a.pakhomov@dell.com"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-06 19:38:20.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-07-06 19:57:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-06 20:28:23.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"CC"},{"updated":"2022-07-06 20:54:29.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-07-06 20:54:56.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-06 21:08:38.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-06 21:28:09.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-07-06 21:31:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-06 22:08:18.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-07-06 22:14:29.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-07-06 22:22:58.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-06 22:53:34.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-07-06 23:30:59.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-07 02:15:11.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-07-07 08:35:51.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-07 08:43:57.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"CC"},{"updated":"2022-07-07 08:44:32.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":34988,"name":"Andrey Pakhomov","email":"a.pakhomov@dell.com"},"state":"REVIEWER"},{"updated":"2022-07-07 08:44:32.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},"state":"REVIEWER"},{"updated":"2022-07-07 08:44:32.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"state":"REVIEWER"},{"updated":"2022-07-12 10:04:13.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-14 09:33:09.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-07-14 12:08:22.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-07-14 15:33:23.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-20 19:13:40.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-07-20 19:58:47.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"REVIEWER"},{"updated":"2022-07-21 18:51:55.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2022-07-22 12:19:13.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-22 15:10:13.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-08-03 14:05:43.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-25 01:31:32.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-08-26 18:39:09.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-08-27 07:19:37.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-08-27 09:42:26.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-08-27 14:47:21.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-01-19 11:36:18.000000000","updated_by":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"reviewer":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"state":"REVIEWER"},{"updated":"2023-01-31 01:17:41.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-01-31 02:24:27.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-05-19 14:27:42.000000000","updated_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"reviewer":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"state":"REVIEWER"},{"updated":"2023-05-23 08:57:54.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-05-28 05:04:20.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2023-06-28 11:10:24.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2023-06-28 19:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-06-29 15:59:18.000000000","updated_by":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"reviewer":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"state":"CC"},{"updated":"2023-06-30 13:26:57.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-12-15 20:55:14.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"cbcdca3ee545856d5aef619cdc2a97a28f96f50a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-07-06 19:31:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86b352c224a921ff3de229c1fab4d5383dc874dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-07-06 19:34:07.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a79785446ae7b64f763c4ede9340ca01ec4da739","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-06 19:38:20.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1657135879806 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"0354415cab1e2498150dd438640f5227e3e6ee56","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-06 19:41:42.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1657136054810 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"249a34be9bd1f8da691c4cfe3476bd019391a6b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 19:57:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bd3e18d2fc04f56beeb390c3c1c56c9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/de6ea1d0fa4e4a639342a50f42639a92 : SUCCESS in 21m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4ee457637f39490f9fa79e0c248db8d6 : SUCCESS in 13m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6f9ac492a4962c2128b152382d1cc4cde1ae577","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2022-07-06 20:28:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"558317f216a60a93dda8fb57e163decbf59b5dc9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-06 20:54:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/848900/2/iscsi/ : SUCCESS in 1h 18m 36s\n- cinder-smb http://cloudbase-ci.com/cinder/848900/2/smb/ : SUCCESS in 1h 19m 44s","accounts_in_message":[],"_revision_number":2},{"id":"02f2f27839d5c3f113d16a852b24d97236e61ef0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-06 20:54:56.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/848900 : FAILURE in 1h 17m 44s","accounts_in_message":[],"_revision_number":2},{"id":"759f404d2c508566a6e5ebe9e6209f8933902f85","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-06 21:08:38.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/848900/2/check/tempest-dsvm-full-focal-py3/dd4a305/ : SUCCESS in 1h 27m 17s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/848900/2/check/devstack-plugin-ceph-dsvm-focal-py3/7e8ee57/ : SUCCESS in 1h 29m 30s","accounts_in_message":[],"_revision_number":2},{"id":"230557ed495d8bca92e1d07dc5229fed0cf995cb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-06 21:28:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/00/848900/2/check/dsvm-tempest-cisco-zonemanager-job/d4418b2/ : FAILURE in 52m 23s","accounts_in_message":[],"_revision_number":2},{"id":"f0c2b8c2471971bf046b6641820ed25375ead0bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 21:31:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5be439fb832a4f95826f04365418bae4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/265a2216135a4ac09e3a00d5298ad1c4 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d49d352de8b45ad8ebef852a9422f8b : SUCCESS in 12m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39ea772840da4b509b2c9195a2c70d50 : SUCCESS in 9m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e30b69a1d63c4599a9d48913fb7d2119 : SUCCESS in 11m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/300ce2c2e4784f8d91285dfc27c6783b : SUCCESS in 11m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/6a18741caecf40b1ba169c4c13a0dee9 : SUCCESS in 51m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1eb26700fabc46148722d6cfcfcace56 : SUCCESS in 1h 12m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f839609dab1c4ff3949ea444fb958f24 : SUCCESS in 44m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cba503fd6aa41bf8b2e021623985921 : SUCCESS in 8m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e0325a19251e461bb978ab52de15ec51 : SUCCESS in 14m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1417a71addec4634a752aa72eef82cbf : SUCCESS in 5m 47s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e96cb6f3686c4cbaa73fa38503afa56e : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/443e5809b6854a5eb1f55c59344e9d5e : SUCCESS in 6m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4f11aea727a44fff90420f05e1225376 : SUCCESS in 6m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7ab116e9af78448fb75efa06ce82f2df : SUCCESS in 38m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d2533a339cc04d2a9b36f80050a80746 : FAILURE in 6m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/309b3bd133924b8aa0687d77402aac19 : SUCCESS in 1h 25m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5656a113fe3a4967bb7dc7a132d0cfb6 : FAILURE in 1h 41m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a2bbfac4a6ba4836b0b81efad8ce09d8 : SUCCESS in 1h 26m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cbb4af61984047fb88e5d5f9da69a6d2 : SUCCESS in 1h 01m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/009610951ce84514a93d8ed04baf3045 : SUCCESS in 1h 14m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cc0af5073f1c4e06a390e46d1a509c09 : SUCCESS in 36m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/aba439bae4dc4ac7a2840ddad9636b51 : SUCCESS in 1h 24m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d30ce3b916ed4113817582e7e010194b : SUCCESS in 1h 54m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d0124880d1f43d2a66c8064d60c9ebe : SUCCESS in 39m 16s","accounts_in_message":[],"_revision_number":2},{"id":"b60d6b8eb4d7b84d69685722a3130e272bb95e6d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-06 22:08:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/00/848900/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/bff0617/ : SUCCESS in 1h 35m 13s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/00/848900/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/205ae75/ : SUCCESS in 1h 40m 36s\n- pure-devstack-cinder-tempest-nvme-fc-aio http://openstack-logs.purestorage.com/00/848900/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-fc-aio/dfc8fa1/ : FAILURE in 29m 08s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/00/848900/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/a0c13ee/ : FAILURE in 28m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"08010e7a3441a53c82c6fdb5b5bf39875a50c702","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-06 22:14:29.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/00/848900/2/check/DellEMC_PowerStore_NFS/c83c4c6/DellEMC_PowerStore_NFS/78 : FAILURE in 42m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"02620a5ea519ecedb87c44fbb15533e70940d346","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-07-06 22:22:58.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/00/848900/2/check/DellEMC_PowerMAX_FC/14d0321/DellEMC_PowerMAX_FC/683 : FAILURE in 40m 09s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/00/848900/2/check/DellEMC_PowerMAX_iSCSI/d8d1bd8/DellEMC_PowerMAX_iSCSI/672 : FAILURE in 50m 59s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"6d13648384940e093aa3554bc0db130f1b93c2e1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-06 22:53:34.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/00/848900/2/check/iscsi-driver-nimble/fadc842/fadc842.html : FAILURE in 1h 41m 13s","accounts_in_message":[],"_revision_number":2},{"id":"a2651d40925fa19b9dc1224b6ecd82d5dfe8d887","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-06 23:30:59.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-00-848900-1/ns5_nfs : SUCCESS in 3h 36m 30s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-00-848900-1/ns5_iscsi : SUCCESS in 3h 36m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"06e03b34411a8cf4d211c01f5131979b9f7ffe98","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-06 23:57:12.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-00-848900-2/ns5_nfs : SUCCESS in 3h 36m 45s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-00-848900-2/ns5_iscsi : SUCCESS in 3h 40m 27s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"e1b6f35aa5a36d756e350168dae565d12c9b7d25","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-07 02:15:11.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/00/848900/2/check/DellEMC_Unity_FC/2758c90/DellEMC_Unity_FC/1106 : SUCCESS in 1h 08m 25s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/00/848900/2/check/DellEMC_Unity_iSCSI/18aad12/DellEMC_Unity_iSCSI/1038 : SUCCESS in 1h 52m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"01b24e08b9fa0e0d5143a6dd8653c7da2c9261f2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-07 08:35:51.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"f02365dc13ec169cd06fc5e978e0e2f8a6c925ed","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2022-07-07 08:43:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b9a91214ffaf1e404f7722243d705980a686b593","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-07-07 08:44:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"41a40d364493efbaa3c01e2683b6de2c71c74620","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-07-07 08:48:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6188bfca82734e19756d080e5d734560e7222a99","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2022-07-07 14:24:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c1c16faca1b6f8c36a23f6c6eacc98528d52415a","author":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"date":"2022-07-07 14:25:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b1fd62363933f80845c43572d331d64d137ae04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-07-12 09:20:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"72cad574651b4c2b6c35d400becdb20a3906bb9a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-12 09:23:39.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1657617630459 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"6b5a85666e7e8d17dae059e43b0b3133e832afac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 09:48:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b6c6983430644c4b405d2cffdf77455\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b67cf0dcca7b41c2862ae45e98484495 : SUCCESS in 23m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a4e15a6fd4744478d726d46a04d83b7 : SUCCESS in 26m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"444a967c043df9e285fe5ee9c62be67e1d8dad46","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-07-12 10:04:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7aa8e6aabebc4bf497a28d247871c9bc : SUCCESS in 42m 35s","accounts_in_message":[],"_revision_number":3},{"id":"7170410fe3c29c19df003585b8400e5ce66aea39","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-12 10:21:18.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/00/848900/3/check/dsvm-tempest-cisco-zonemanager-job/e2ad7fd/ : FAILURE in 52m 20s","accounts_in_message":[],"_revision_number":3},{"id":"4ee5fbf430fd36d93dcfc77ea7e7c86f492672c8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-12 10:29:24.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/848900/3/iscsi/ : SUCCESS in 1h 07m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/848900/3/smb/ : SUCCESS in 1h 08m 25s","accounts_in_message":[],"_revision_number":3},{"id":"ca9f91b8803a80534165ccb31cb881e4e7eeaf51","author":{"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},"date":"2022-07-12 11:26:26.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0e4e643b33d2a87861fbd3707d7696677fccb948","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-12 11:49:06.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/00/848900/3/check/iscsi-driver-nimble/7afd38c/7afd38c.html : FAILURE in 1h 41m 59s","accounts_in_message":[],"_revision_number":3},{"id":"0574774a702a77dcb015a3280eddea3c4b352a45","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-12 12:03:54.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/00/848900/3/check/DellEMC_Unity_FC/6df9ac4/DellEMC_Unity_FC/1133 : SUCCESS in 1h 11m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/00/848900/3/check/DellEMC_Unity_iSCSI/955315d/DellEMC_Unity_iSCSI/1064 : SUCCESS in 2h 00m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"b5312e1d802e8a942056b2bdc2c3a82cf5b7068b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-12 12:04:34.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/848900/3/check/tempest-dsvm-full-focal-py3/2fda3c8/ : FAILURE in 2h 40m 17s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"c7341527c47b3ae51445c992512375081ea5ab9c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-12 12:53:39.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-00-848900-3/ns5_nfs : SUCCESS in 3h 29m 14s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-00-848900-3/ns5_iscsi : SUCCESS in 3h 26m 49s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"fe446c4c34514075e22cf5c863006b492931f628","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 13:38:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8d72bb5cc864e4caf03e85ee655ce5c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b97c9ac5ca454b9b9b5dd0bd6eea49e1 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db56e7ad8f1446fdae9fff9f3f19e480 : SUCCESS in 12m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cba7a6dc30f54fb08b0118f43081b251 : SUCCESS in 11m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33b9faf2db7f4925a8c9e07ddc5e481f : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86c7cd286d1c486891745653988e4d62 : SUCCESS in 12m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/558f045a4b9c46fdaaa3f85c30f66697 : SUCCESS in 52m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f54bcab3ea064a53bee8b6e08064ef27 : SUCCESS in 1h 20m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16d0089a7c4b4c08ad2f2336b825a739 : SUCCESS in 44m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3aa6c1ba76124d259d9f4cf9a27380a3 : SUCCESS in 8m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4b21adb6dff64de4b52fa8f312de00cd : SUCCESS in 15m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/372f73ed9214490ca8c4120d0210c69b : SUCCESS in 6m 13s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/77e1c734b3864d59b00db43e5f36f2cf : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9be622de25f146bab265607d0464b2d0 : SUCCESS in 7m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/00479be79a9246e7bf7dbc867ee220ed : SUCCESS in 6m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/66707cffafd1443283c179cc8e249536 : SUCCESS in 41m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bf84ec8be3be4318acff5a54f7cf6952 : FAILURE in 6m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a41093fc3007449c9150d97244fabbd4 : SUCCESS in 1h 32m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2d6f73d3d2274eaead38af06bab6fd15 : FAILURE in 1h 38m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bd9276911baa40ea98f72b2acb3245ed : SUCCESS in 1h 35m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bb9309e38a994625a48171263852ac37 : SUCCESS in 1h 00m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0752faf3679547aca5b1e864e95d1a81 : SUCCESS in 1h 33m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d21c12d3f4754221b46aa61679e28872 : SUCCESS in 37m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8870056bd1434b2c9c97c6d67e8f7ae4 : SUCCESS in 1h 29m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5f635781bbd540979b872d8cb1ac8f38 : POST_FAILURE in 2h 17m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/229449c08487490fab14c233e481ed6e : SUCCESS in 53m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b8cc4ae1a273a2b83ef49fe3506e300d13790634","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-13 02:36:29.000000000","message":"Patch Set 3:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":3},{"id":"c015853eef24e3c0ea54c99e37b7464adba1ddd4","author":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"date":"2022-07-14 09:26:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ef98fb096a9c4db16e24b46fd2d05f389ca2e4d1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 09:33:09.000000000","message":"Patch Set 3:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/56","accounts_in_message":[],"_revision_number":3},{"id":"fdcf383962ddeab8c5717b915e82542cc53a6b1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 09:53:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdfe814da60f4e12a554bf68d30e8c32\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38c98d2812d645fb8102ce9f505799ab : SUCCESS in 25m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6560ef13473b41a986c53063176c61ff : SUCCESS in 22m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"599b3dd99986b598bcfb218973bbe80d48503c8f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-07-14 10:26:02.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d553ebf4ab104ba5be5a6e9e5792bbe4 : SUCCESS in 43m 26s","accounts_in_message":[],"_revision_number":3},{"id":"6c7db8031cba488ac23bebc1e3d177d57a544170","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-14 10:39:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/848900/3/iscsi/ : SUCCESS in 58m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/848900/3/smb/ : SUCCESS in 1h 12m 07s","accounts_in_message":[],"_revision_number":3},{"id":"068b3de5014c521f4a3af0b9fddb3e64e97f7aa0","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-07-14 11:22:30.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d5484210616e49c516dff95e568d4e6afe80fcd4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 11:28:24.000000000","message":"Patch Set 3:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/69","accounts_in_message":[],"_revision_number":3},{"id":"e8d4e747516e361690b5512f7db4b23ec0c52758","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-14 11:34:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/00/848900/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/0ba5ee5/ : SUCCESS in 1h 27m 57s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/00/848900/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/429c072/ : SUCCESS in 1h 28m 44s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/00/848900/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/aa23b42/ : FAILURE in 27m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e0c86acc798bdf9eed9a0ab4e7479de5788f79ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 11:54:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebd98f853554406e8b95f37eed0c472a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/60da0bcd82924a708a8d9fd5958748e3 : FAILURE in 17m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0da30ac4b18c441d8bc8d7f432e0fe33 : SUCCESS in 29m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"91c488ec3669e071f18b62544dbd140763cf7a87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 11:56:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09657650076846c0a860e9a41588105e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f08fbee48fdd427b9b8505a91f61f2b1 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1deeb487d57b47b98062ecd6f058de43 : SUCCESS in 12m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a4cf83cb71c4fde81f098627587fbd5 : SUCCESS in 12m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5530c7a22e448ed86b49be4fa2d3634 : SUCCESS in 12m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3676de43b021460b97f8fc7e1ac138ba : SUCCESS in 12m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/86353c77b56b47bea5089d5adefaf8ae : SUCCESS in 1h 01m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d7f5102d0b7b4fd0ad719892430d7256 : SUCCESS in 1h 28m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c56600cb3bc1408f9bf7d29234d2849e : SUCCESS in 52m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7231b282ab04206a071fb4405afb55d : SUCCESS in 10m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b6a021a77b7347519c793de4dc3beffa : SUCCESS in 17m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/df57ab29118b4a5ab182599b57398a91 : SUCCESS in 6m 21s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/419a9ab8c1584d13965f0de9e2e53cbb : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b43110a4b9804f8d9366089dbc025f37 : SUCCESS in 6m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a149a2dfc8db4aae81be483dbf1d6fde : SUCCESS in 7m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9723774f339d43fd8123d8001f55a785 : SUCCESS in 41m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/12d2cdcea78349f48e4083f88fac94d1 : FAILURE in 7m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/db9a6148b3ad452389d5e3660c58b23e : SUCCESS in 1h 26m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/efad6b6dbf9e4919aa6350fea3c858a1 : FAILURE in 1h 45m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/94d3728591ba4cf89471a62a221c09d7 : SUCCESS in 1h 36m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/78f2db51b4604c4284badcc3ae58b65e : SUCCESS in 1h 14m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/63c05fe707b948b986c284e45aec01ee : SUCCESS in 1h 28m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/39b75d6d2bcf4a79a878dae2a33bf6a8 : SUCCESS in 52m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/42fdaf2d1922471382f5557b256686e3 : SUCCESS in 1h 30m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cfc3b31b381a44539a5e50aa1feb4989 : SUCCESS in 2h 28m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64c6aa579fa34873bb6e8074446cbf1a : SUCCESS in 1h 02m 27s","accounts_in_message":[],"_revision_number":3},{"id":"b3850ddb48972909b843de7413e13a988a1af49d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-14 12:08:22.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/848900/3/upstream-check/cinder-tempest-plugin-ontap-fcp/2ab8551/ : RETRY_LIMIT in 9m 33s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/848900/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/e174e52/ : RETRY_LIMIT in 8m 17s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/848900/3/upstream-check/cinder-tempest-plugin-ontap-nfs/7c43ef4/ : RETRY_LIMIT in 8m 36s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"5d8d13750f5bba959fb2339d4a64c3fc0a00022d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-14 12:14:32.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/848900/3/check/tempest-dsvm-full-focal-py3/35aea16/ : SUCCESS in 1h 46m 35s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/848900/3/check/devstack-plugin-ceph-dsvm-focal-py3/451223a/ : SUCCESS in 2h 44m 13s","accounts_in_message":[],"_revision_number":3},{"id":"9788edd51066a0dbe095d153db46dfbf985090c5","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-14 12:16:30.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/00/848900/3/check/DellEMC_PowerStore_iSCSI/fddb086/DellEMC_PowerStore_iSCSI/602 : ABORTED in 9s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/00/848900/3/check/DellEMC_PowerStore_FC/9734040/DellEMC_PowerStore_FC/641 : ABORTED in 20s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"c1d44d1a056c6447c9adede1094948a40d54fe69","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 12:16:31.000000000","message":"Patch Set 3:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":3},{"id":"099289a732bdd4eb8caba25b3daf4e4198c2559c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-14 12:25:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/00/848900/3/check/dsvm-tempest-cisco-zonemanager-job/d8909f5/ : FAILURE in 51m 57s","accounts_in_message":[],"_revision_number":3},{"id":"278b891cef87bfa79df1c96e0068e91397d0700a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-14 12:32:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/848900/3/iscsi/ : SUCCESS in 1h 09m 30s\n- cinder-smb http://cloudbase-ci.com/cinder/848900/3/smb/ : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":3},{"id":"f617938701472e78002091d29a2a6a02c381c2de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 13:03:25.000000000","message":"Patch Set 3:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/77","accounts_in_message":[],"_revision_number":3},{"id":"f0856eb9310768bee6bb59089cec3ad1ce0724b8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-07-14 14:45:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/926d94d7dd0a4d8f81e6b4d47465269a : SUCCESS in 45m 01s","accounts_in_message":[],"_revision_number":3},{"id":"0cf49061ca5c67490645a2b5cd8bfc32e049ff58","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-14 15:33:23.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/00/848900/3/check/DellEMC_PowerFlex/d5fe76c/DellEMC_PowerFlex/990 : FAILURE in 1h 43m 31s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"438c8474d239d5d150284320e2a39f8fca20c496","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-20 19:13:40.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/3/20220721020017/NEC-V-FC : SUCCESS in 1h 43m 20s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/3/20220721020017/NEC-V-ISCSI : SUCCESS in 2h 13m 20s","accounts_in_message":[],"_revision_number":3},{"id":"dc2917cd5529f53fa4633fb16489e5ae17f89c75","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2022-07-20 19:58:47.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0831b32ff23648814012906c6d9c62bb1795791d","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-07-21 18:51:55.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"acc39f7c208b15e0c0b0f2de36173e0f35e429b8","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-07-21 18:52:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7fa6b364710d6858925dedb35c31075692c30dfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-07-22 10:30:04.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"2ede3dca65dc20bbe82de88143473c6b057d08e6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-22 10:34:06.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"2613b5596ddbfff9f0efa7e7294fb7b8e4b392f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 10:52:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/201da416286e492495306d9c79076fcf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77285af2ba9849d99840d33b60ae52c3 : SUCCESS in 20m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e2c1c9f558b54f32919fbc38cf2cfb73 : SUCCESS in 15m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"371b72863a2d3ad99ecc56b30f609d4a69e025fc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-22 10:57:24.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/00/848900/4/check/iscsi-driver-alletra-6k/ca2a504/ca2a504.html : FAILURE in 25m 39s","accounts_in_message":[],"_revision_number":4},{"id":"5573f8e4c469f9539a33bb7eaba835b6547e3355","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-22 11:29:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/848900/4/iscsi/ : SUCCESS in 57m 48s\n- cinder-smb http://cloudbase-ci.com/cinder/848900/4/smb/ : SUCCESS in 58m 08s","accounts_in_message":[],"_revision_number":4},{"id":"570afa271cf8cdb83431c35b82c69df2be7f8287","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-22 11:30:02.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/848900 : FAILURE in 54m 12s","accounts_in_message":[],"_revision_number":4},{"id":"98d7faa920d62ec981201fe7ef405f1bad64bfba","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-22 12:05:04.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/848900/4/check/tempest-dsvm-full-focal-py3/86e366c/ : SUCCESS in 1h 22m 34s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/848900/4/check/devstack-plugin-ceph-dsvm-focal-py3/5a768f7/ : SUCCESS in 1h 31m 07s","accounts_in_message":[],"_revision_number":4},{"id":"b2a1c6ac67e81fa452ca8804c6db0655d75d1a08","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-22 12:08:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/00/848900/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/6a502a1/ : SUCCESS in 1h 35m 17s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/00/848900/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f755d8b/ : SUCCESS in 1h 30m 26s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/00/848900/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/3a474dc/ : FAILURE in 27m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"82cd173cdab40348f2861411a2d24b95d1270c48","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-22 12:16:46.000000000","message":"Patch Set 4:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/220","accounts_in_message":[],"_revision_number":4},{"id":"e01d1095858cd113a37be5373c7656522a4ca6d7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-22 12:19:13.000000000","message":"Patch Set 4:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/00/848900/4/check/fujitsu-eternus-dx-iscsi/c39cfdb : SUCCESS in 1h 44m 51s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/00/848900/4/check/fujitsu-eternus-dx-fc/c0a44da : SUCCESS in 1h 47m 24s","accounts_in_message":[],"_revision_number":4},{"id":"f39fbcf26c2ef910c5e4d7a1a4f5937aef670910","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 12:57:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbbb15ae22624d009cb0af679eb55135\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3222f3895534a0d84da9564b4fa3061 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4ed02841d4e4b6fb1738db337403462 : SUCCESS in 10m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd402643f95b43fdbf11a35973d9b114 : SUCCESS in 11m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67c08b2e91e74e7eade935ba8897fbb7 : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e7ac98072884bc7b437025d1915c1fb : SUCCESS in 13m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/1eee40295bd94bbfabeb57e0efbe6324 : SUCCESS in 1h 00m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fbf216fe49e9475ba755ede6e55bddba : SUCCESS in 1h 06m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1179d37d57224115ba5929b4a36b297e : SUCCESS in 56m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52bde1fdfb04462aa752190a74178d9b : SUCCESS in 8m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/adee2ef84d5840d2b0e5899e6baded02 : SUCCESS in 15m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/eb3425d5a83a4492b5145c947242617b : FAILURE in 6m 42s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0a4420475d2e4a83a09102499569754e : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bc8905b5e4074787a46fd49a79085918 : SUCCESS in 6m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cbcd3591835e4d7597c8ce064534684c : SUCCESS in 8m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b4bc68873ee54a90b299ffaf22b2f619 : SUCCESS in 46m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c0a5166c3ed643fcae3f3e6cfa782a90 : FAILURE in 6m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f38ff7c427224a3bafd4f39d21c6c45c : SUCCESS in 1h 38m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/344e19252bdc484890cc223445cb3238 : FAILURE in 1h 45m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8899555a22d34e86b6ab43dc00122937 : SUCCESS in 1h 37m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5faad6962ae64d0e8f83d2b47cf9b164 : SUCCESS in 1h 01m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/eb39f7997dfc4deeabf3aac6f06038e6 : SUCCESS in 1h 20m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ae80b91db45d4354b2f515c94454976b : SUCCESS in 48m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fbfd2570281c4a158cb1ee4b8088cd1e : SUCCESS in 1h 28m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6c21283f89e647beb11f1026bfff1b69 : SUCCESS in 2h 07m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f05d5df81d546db8d583c6198a48ab9 : SUCCESS in 53m 47s","accounts_in_message":[],"_revision_number":4},{"id":"777b0717e89777d9d452ef59d92991c2f3b4d724","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-22 13:49:06.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/848900/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/ddea2f2/ : FAILURE in 50m 15s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/848900/4/upstream-check/cinder-tempest-plugin-ontap-nfs/6bccaab/ : RETRY_LIMIT in 7m 27s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"3f92cf3ed3a0901b3e7f09dd9d82e2553a12f4f4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-22 14:33:22.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/00/848900/4/check/DellEMC_PowerFlex/4767245/DellEMC_PowerFlex/1057 : FAILURE in 1h 35m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"564af9a8f8c39e66bffab22b4a17f1b7e9f8a943","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-22 14:36:38.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-00-848900-4/ns5_nfs : FAILURE in 43m 55s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-00-848900-4/ns5_iscsi : SUCCESS in 3h 43m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"2b1adefccbaa14042d94dc57a6e878b8c668b13b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-07-22 15:10:13.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/00/848900/4/check/DellEMC_SC_FC/c93d073/DellEMC_SC_FC/790 : SUCCESS in 49m 33s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/00/848900/4/check/DellEMC_SC_iSCSI/d1283a1/DellEMC_SC_iSCSI/813 : SUCCESS in 2h 12m 04s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"0bc4abbf495f53e5ed60f8dc7a0db30c8b73feaa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-22 18:32:50.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/00/848900/4/check/DellEMC_Unity_FC/24aedbe/DellEMC_Unity_FC/1233 : SUCCESS in 1h 20m 12s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/00/848900/4/check/DellEMC_Unity_iSCSI/3b94869/DellEMC_Unity_iSCSI/1137 : SUCCESS in 1h 47m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"129433ad14b29c079a2375ce1001272ec30f2fa5","author":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"date":"2022-07-25 12:58:57.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"74397d17edc92a991d5cda2506a1e5fabed7868e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-28 05:42:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/4/20220728122638/NEC-V-FC : SUCCESS in 2h 00m 00s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/4/20220728122638/NEC-V-ISCSI : SUCCESS in 2h 16m 00s","accounts_in_message":[],"_revision_number":4},{"id":"1cbbe69fc7ea0994b57b961a236ae554b337a14d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-03 14:05:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/4/20220803193223/NEC-FC : SUCCESS in 3h 25m 29s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/4/20220803193223/NEC-ISCSI : SUCCESS in 3h 33m 17s","accounts_in_message":[],"_revision_number":4},{"id":"40c94321c9c86a6cc833074b2dd8d041c40c9212","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-24 09:22:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bbafda2cf11e0494a2ee847caa73215a70438852","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 09:45:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5b42958859a4437b336f46077a4c2c6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a4ff553f71d4c76af2477e96e653433 : SUCCESS in 16m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d23245098224e5985688a9832fb9692 : SUCCESS in 20m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fc3336c38fe56adf6aa31e6aa69e5bec0d346d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 11:36:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b3a2979693f44bcb98245e59449db63\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40fe523da07f49e59ca59cf0478c2813 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b8588918a45415b8f77a449fc8fbfae : FAILURE in 17m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3176de54297f42e3b9a217f0b8e33c03 : SUCCESS in 12m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/035a145715de40c6ba114d589fe2c58d : SUCCESS in 13m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01c0bfb294c64810a531cbfe73b520fe : SUCCESS in 13m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/f26a1f12363a475a92dd9c97c857e0c4 : SUCCESS in 1h 05m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9c8af4410cd744e4a70a770f93a69bde : SUCCESS in 1h 26m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36438de89a094d9db481352f4af5ebe4 : SUCCESS in 48m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3412032ce9d94d3cb0ff7b2af4ec664a : SUCCESS in 7m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9315553189bf48fc8026efd3928e7868 : SUCCESS in 15m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3335fad6bbf8474ea57f0b94695c667a : SUCCESS in 4m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eded26ee5c9047d882e98503a332d21c : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/28493563d79d48eb9f42456c58aa63c6 : SUCCESS in 5m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/188d5bad68e54dbc89c9c4365802e61d : SUCCESS in 7m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8e5695fa567743c998872e1dfc4efe7a : SUCCESS in 39m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4b910eccdf374bc69f18f78fb8a59ffa : FAILURE in 7m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a870b10742ea4aef862a4098bdbd2164 : SUCCESS in 1h 10m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/14b6688c976743af9a83f570ea243ca9 : FAILURE in 1h 23m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4822793e1d314afcb37284dde29b6f97 : SUCCESS in 1h 41m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/52ce6fe2d6874bd095652c25f3660f74 : SUCCESS in 1h 12m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e2269ad8bcce4765a05ad01dca047220 : SUCCESS in 1h 17m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ca01c08e094440de82c680e6f5bd9eb9 : SUCCESS in 39m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e90ebaf9c90f4640a40440058998da9d : SUCCESS in 1h 25m 44s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b109be3a5cf84b088fc2e4ce41b51db3 : SUCCESS in 2h 07m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9e83e6b616c4398a144c348caf9b739 : SUCCESS in 55m 21s","accounts_in_message":[],"_revision_number":5},{"id":"203e7fb584dbd5a7f6d649a90cb7600e742fbe5c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-24 12:46:06.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/848900/5/iscsi/ : SUCCESS in 53m 30s\n- cinder-smb http://cloudbase-ci.com/cinder/848900/5/smb/ : SUCCESS in 1h 10m 39s","accounts_in_message":[],"_revision_number":5},{"id":"ff0cda1156a1b31f4f8283c2cc71fb9fa888b3ba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-24 13:03:58.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661332974355 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"435612dc2a51e4c55eac7b95bc246b7c4a60c898","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-24 13:23:51.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/848900 : FAILURE in 1h 03m 04s","accounts_in_message":[],"_revision_number":5},{"id":"f244f63f9ef1c3befd6821fc7f2991f94c135f3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-24 13:49:49.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"40f39521f722b17de1fc6fc4a9e9d3dc67f7c0ce","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-24 13:49:51.000000000","message":"Patch Set 5:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":5},{"id":"fc9553caac188e54d892743244b89a9d887996a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 15:20:43.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc78cad780284c358754c253179b878f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3030ecb6185c4d1a8b2a07ad456bc7f4 : SUCCESS in 21m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6eccb7f3c6884c738cb80ffddfa0c667 : SUCCESS in 20m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2de412dccf83d67a7e767d37c9d8d8595e0aad29","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-24 15:59:59.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/848900/6/iscsi/ : FAILURE in 1h 14m 08s\n- cinder-smb http://cloudbase-ci.com/cinder/848900/6/smb/ : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":6},{"id":"7831e151045ab4063cbd0b0bb23e09b37dbbe681","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-24 16:02:53.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-00-848900-5/ns5_nfs : SUCCESS in 3h 35m 34s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-00-848900-5/ns5_iscsi : SUCCESS in 3h 32m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"cbf62c13c63a8a92ddab8582965739aead91eac9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-24 16:10:55.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661349000482 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"1240fa6928ccfcf9fac1472b35efd8d68faf1be9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 16:22:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6396b033716440a3a97ddd3224e4e5ea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e9c48cd2cd64285a264a0ee3d77fd10 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eab7494fad364787a1caef30c1ecd9d1 : SUCCESS in 11m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec84faf36a434f449d23bf793c1c07bf : SUCCESS in 11m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77684dfe7aa94b6fa08918f6f272a00a : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd823ac70bcb427c96b04adc4ab7df42 : SUCCESS in 12m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/a1efad2459c446e68907a17aebb81d70 : SUCCESS in 58m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6c8c36d7632b43a0b838ce5a3ae421ab : SUCCESS in 1h 26m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba79642a4475459995d717edc5a4b9ee : SUCCESS in 50m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25404058275243398b98ba929a831dd9 : SUCCESS in 9m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dc17d3a37104439cba98b4f3f421f491 : SUCCESS in 13m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/348d807cbdf9439cb735f54844d53282 : SUCCESS in 7m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/03ec835982ee4d26ad65628a722d85f0 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aadcc03ea58c4a528dfd22f5d15917d6 : SUCCESS in 6m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/56816751f8c54f7d8315878107cd27b5 : SUCCESS in 6m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5a881a199a7d449f9d69ce566672891e : SUCCESS in 41m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c3e537a3a022486188c81abb3886891f : FAILURE in 7m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4587e5ff74434934af9d370b8b0f1ce0 : SUCCESS in 1h 23m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fac2ded3a5ce49fe9d7047665c3aec95 : FAILURE in 1h 39m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/894d1fb14862480e8dec119a05c845b3 : SUCCESS in 1h 46m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3252a05368f94e0eb005091655fa1b20 : SUCCESS in 1h 19m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/af93341e2c874c2e944e936f160cdf3c : SUCCESS in 1h 42m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/752153a938c94b89954f6996c6d301c3 : SUCCESS in 58m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bb591d27811944b3a4c47acf8d3dd703 : SUCCESS in 1h 44m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bfa937a287bb4219ae22266b0574fe3c : SUCCESS in 2h 24m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/695758e9f4354deda6008ad56fee7964 : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":6},{"id":"c50680523a5c23e5f92006cf62564c496cfa113d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-24 17:28:47.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/00/848900/6/check/dsvm-tempest-cisco-zonemanager-job/ae71a9d/ : POST_FAILURE in 13m 14s","accounts_in_message":[],"_revision_number":6},{"id":"4bd080a9f47741f705c4e05a7553c8065c3f24fd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-24 17:29:03.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/848900/6/check/tempest-dsvm-full-focal-py3/08a9026/ : SUCCESS in 1h 30m 11s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/848900/6/check/devstack-plugin-ceph-dsvm-focal-py3/2a9511b/ : SUCCESS in 2h 06m 37s","accounts_in_message":[],"_revision_number":6},{"id":"d2c6634a9386ebbd2ea85f9c8ba70770e5530266","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-24 18:01:08.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/848900/6/upstream-check/cinder-tempest-plugin-ontap-fcp/3c44837/ : RETRY_LIMIT in 8m 00s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/848900/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/2ccb785/ : SUCCESS in 1h 24m 25s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/848900/6/upstream-check/cinder-tempest-plugin-ontap-nfs/4b1a2b8/ : SUCCESS in 1h 21m 01s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/848900/6/upstream-check/cinder-tempest-plugin-solidfire-iscsi/f32843e/ : SUCCESS in 1h 36m 50s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"0a91b16058db804a734217e473a8b330b42a834f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-24 18:28:35.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/848900 : FAILURE in 1h 01m 44s","accounts_in_message":[],"_revision_number":6},{"id":"e76f48909770dbf411777c91ac09268fed76826c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-24 19:01:22.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-00-848900-6/ns5_nfs : SUCCESS in 3h 33m 34s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-00-848900-6/ns5_iscsi : SUCCESS in 3h 32m 31s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"c59f08193b6191a8b01d8a70365ac3f907938784","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-24 20:23:42.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/00/848900/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b275e59/ : SUCCESS in 1h 50m 11s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/00/848900/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d013709/ : SUCCESS in 1h 54m 36s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/00/848900/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/4af5253/ : SUCCESS in 1h 31m 15s","accounts_in_message":[],"_revision_number":6},{"id":"b80208b89387fec4764f217e4a585dc701bc26bb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-24 21:46:13.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/00/848900/6/check/iscsi-driver-alletra-6k/ec180a8/ec180a8.html : FAILURE in 1h 02m 18s","accounts_in_message":[],"_revision_number":6},{"id":"013a6e640bc8be39024aa17a449e3cad314c8ed5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-24 22:12:07.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/848900 : SUCCESS in 1h 28m 47s","accounts_in_message":[],"_revision_number":6},{"id":"72dd7334e163ae5e6233feecf9d11047713fb8c0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-25 01:31:32.000000000","message":"Patch Set 6:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/25_08_2022_01_07_50\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"f3374b1617f2ec8415366abd4b873da682859179","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-25 03:32:32.000000000","message":"Patch Set 6:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/00/848900/6/check/fujitsu-eternus-dx-iscsi/96bc4ba : SUCCESS in 1h 45m 27s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/00/848900/6/check/fujitsu-eternus-dx-fc/3a75c4d : SUCCESS in 1h 26m 02s","accounts_in_message":[],"_revision_number":6},{"id":"826f057796942501aa0430014e834fc1db1ea951","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-25 21:13:41.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/00/848900/6/check/DellEMC_PowerFlex/200d0fd/DellEMC_PowerFlex/1219 : ABORTED in 2h 00m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"971ccc4b37cf9cc2da85478679c4a55d7b00cb93","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-26 02:54:29.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/00/848900/6/check/DellEMC_PowerStore_FC/e2836aa/DellEMC_PowerStore_FC/787 : SUCCESS in 1h 20m 59s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/00/848900/6/check/DellEMC_PowerStore_NFS/776eaf8/DellEMC_PowerStore_NFS/327 : SUCCESS in 1h 29m 24s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/00/848900/6/check/DellEMC_PowerStore_NVMeTCP/4ae7c65/DellEMC_PowerStore_NVMeTCP/202 : SUCCESS in 1h 47m 21s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"526e8f5ab44d31660639586b3767c2c576a017af","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-26 14:14:53.000000000","message":"Patch Set 6:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/487","accounts_in_message":[],"_revision_number":6},{"id":"4dca3210329ecd231f27dc1ac8cdc42b3f7a9322","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-08-26 18:39:09.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/848900-6/HPMSA-iSCSI/ : SUCCESS in 56:29.04\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/848900-6/HPMSA-FC/ : SUCCESS in 53:33.20","accounts_in_message":[],"_revision_number":6},{"id":"4f9d623759a3eef8d5184e2d330bbd6af6b4b906","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-26 23:53:33.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/00/848900/6/check/DellEMC_Unity_FC/5548786/DellEMC_Unity_FC/1564 : FAILURE in 14m 14s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/00/848900/6/check/DellEMC_Unity_iSCSI/82d58c9/DellEMC_Unity_iSCSI/1439 : SUCCESS in 1h 56m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"98cc56c23d6e7ab6af6af084b86dc1784518ca0a","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-08-27 07:19:37.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/848900-6/PVME-iSCSI/ : SUCCESS in 1:07:33\n- PVME-FC https://os-logs.seagate.com/dell-ci/848900-6/PVME-FC/ : SUCCESS in 1:04:34","accounts_in_message":[],"_revision_number":6},{"id":"b4849c6932f940aea6f509289a63bb70d86395ee","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-08-27 09:42:26.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/848900-6/Lenovo-iSCSI/ : FAILURE in 1:18:02\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/848900-6/Lenovo-FC/ : SUCCESS in 1:04:43","accounts_in_message":[],"_revision_number":6},{"id":"f6a16306c0a7bef7defeb5e88d389678f35ac2a3","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-08-27 14:47:21.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/848900-6/Seagate-iSCSI/ : SUCCESS in 1:10:43\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/848900-6/Seagate-FC/ : SUCCESS in 1:07:06","accounts_in_message":[],"_revision_number":6},{"id":"5cc096915a145a03e7bb632854140c318e70a238","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:38:26.000000000","message":"Patch Set 5:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":5},{"id":"125bc8b21712ac75e3de7d657aafa1acd2056b55","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-30 05:52:19.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/6/20220830111756/NEC-FC : SUCCESS in 3h 32m 00s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/6/20220830111756/NEC-ISCSI : SUCCESS in 3h 34m 20s","accounts_in_message":[],"_revision_number":6},{"id":"7a5efba680f12c02b40b4ff3448ab56c901a7d1e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-09-05 16:34:05.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/6/20220905190054/NEC-V-ISCSI : SUCCESS in 3h 05m 58s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/6/20220905190054/NEC-V-FC : SUCCESS in 6h 33m 07s","accounts_in_message":[],"_revision_number":6},{"id":"6e703fe68e7197602440e4969997ea8ae3522fe7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-09-14 23:13:29.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1ba0f7e08d3441fb82d0affd89b84cc7 : SUCCESS in 39m 05s","accounts_in_message":[],"_revision_number":6},{"id":"f9a5a982fe4e04b99a41861d241d30fe87321293","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2022-12-19 10:42:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"048487f11a893743c54a289d330f5475568c6685","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-01-10 08:03:41.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"b7a724becbc85f7ab124367fd3c44616f1d4210e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-10 08:05:08.000000000","message":"Patch Set 7:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2ae5fcc1e976bc93e68cfa577fc1e8e3d65a64c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 09:47:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da3cd00b73d9492a8b58adb71a58e254\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/141f4aef753d46aa89e18e24770655cc : SUCCESS in 22m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/971b80a255594a6aad397db5b485a56b : SUCCESS in 21m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9c3071b603480c4ae421744a2c3ac5374feb86a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 10:04:57.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/973c3d380ecb4825a947b37df5445ec3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bba4748d97784573b8ee3a5f35473012 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/921c46433a0944a2a51a11f856ec1111 : SUCCESS in 10m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/703649e203d24bcca946dba9eee8a6b7 : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96565c9f90ed4f228479763068d58035 : SUCCESS in 13m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/bd10608a7af144d98c9cb6e5e238570c : SUCCESS in 49m 25s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/34b46d877abe4cba876406bb3e24691c : SUCCESS in 1h 11m 08s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2afa3d7babd84f35bdb8bcf4f4e16454 : SUCCESS in 1h 13m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f6f9c2a2c644281921c95517b647efd : SUCCESS in 1h 00m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecd179b4532b4c15901c05e3bcfc5bf6 : SUCCESS in 7m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/583cc68aa221488ca289657e758a0301 : SUCCESS in 15m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2b1cce85f06b4be7a785ec66de1689ca : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a8a86d0bc6de4c99ba5abcef93c67f8b : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/051c011d7ebd4759a6b3059880d613f7 : SUCCESS in 5m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5755c9ab632b48c591642433b3bbb163 : SUCCESS in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c0e9b605886e4d16901104398ab486ef : FAILURE in 18m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/42cdc8c3b17f482ba169c95570392acf : FAILURE in 4m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0330748038a94735b9fb71fec984ca75 : SUCCESS in 1h 04m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c85d16d4fb00489f8b05d4b6beb96004 : FAILURE in 1h 24m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5ffc2ab7ded54dbc999243f6c63a9b62 : SUCCESS in 1h 37m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ffdf52bd5e1d41fa998b951e2d7c507a : SUCCESS in 57m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d4744b46c2e7406a9f970a88476f5285 : SUCCESS in 1h 36m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8bd735b02c12449c8638f55a3c013674 : SUCCESS in 43m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8c925ce4613b4d5cb18eeed3780cc450 : SUCCESS in 1h 45m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/802745505504436e8416ed57fe2b512f : SUCCESS in 1h 57m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a3bf89090fd45c3b642f1593ac3e6e1 : SUCCESS in 41m 25s","accounts_in_message":[],"_revision_number":7},{"id":"e62b81d1900bed4bf36f0b8a5bab2f53b6d20cfb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-10 10:05:54.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/00/848900/7/check/iscsi-driver-alletra-6k/4bff34e/4bff34e.html : FAILURE in 1h 10m 16s","accounts_in_message":[],"_revision_number":7},{"id":"a4ea2ce6ae9a744d995353e285afce467baa0a1b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-10 10:30:43.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/848900 : SUCCESS in 1h 26m 51s","accounts_in_message":[],"_revision_number":7},{"id":"07c8b217ccd189e12d12e0d23d03ce24537c458c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-10 10:30:47.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/848900 : SUCCESS in 1h 26m 13s","accounts_in_message":[],"_revision_number":7},{"id":"ea7242af815bfd66108617e47607d7962488e272","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-10 11:04:28.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/848900/7/upstream-check/cinder-tempest-plugin-ontap-fcp/09d20e3/ : SUCCESS in 58m 03s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/848900/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/8b6aa52/ : FAILURE in 40m 37s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/848900/7/upstream-check/cinder-tempest-plugin-ontap-nfs/9d6a831/ : SUCCESS in 43m 18s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"d69e1aa5b2f0ac5cf19a19c8bcffbdf87f624008","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-10 11:13:16.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/00/848900/7/check/iscsi-driver-nimble/27922a1/27922a1.html : FAILURE in 1h 37m 26s","accounts_in_message":[],"_revision_number":7},{"id":"389b5ce3474a6e802e83ba54853909412cbd0dcd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-10 11:17:10.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/00/848900/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/55d88e4/ : SUCCESS in 1h 50m 47s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/00/848900/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/ce97397/ : SUCCESS in 1h 45m 18s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/00/848900/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/513b349/ : FAILURE in 2h 14m 50s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/00/848900/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/389feaa/ : SUCCESS in 1h 51m 07s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0398883c94818e7e16460c3ffcfe76b7fb6b1df8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-01-10 11:33:19.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/00/848900/7/check/DellEMC_PowerFlex/16191a0/DellEMC_PowerFlex/1508 : SUCCESS in 1h 28m 00s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"b0f14c89677f0a39877f3dd404ab922c28639a42","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-10 11:57:10.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-00-848900-7/ns5_nfs : SUCCESS in 3h 24m 15s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-00-848900-7/ns5_iscsi : SUCCESS in 3h 30m 24s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"819fa6229058b75421c4e3b1e1ede37d846f0720","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-10 14:41:03.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/00/848900/7/check/DellEMC_Unity_FC/8d439ac/DellEMC_Unity_FC/2123 : SUCCESS in 1h 40m 31s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/00/848900/7/check/DellEMC_Unity_iSCSI/d5148a3/DellEMC_Unity_iSCSI/1994 : FAILURE in 2h 45m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"1439ff1f24e1e8a3dd8eb9fa483feaece7450ecb","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-01-11 01:07:44.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/7/20230111031916/NEC-V-FC : SUCCESS in 6h 06m 35s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/7/20230111031916/NEC-V-ISCSI : SUCCESS in 6h 48m 25s","accounts_in_message":[],"_revision_number":7},{"id":"7af3105dd00e06d0befb881a5a93b87e27d859cb","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-01-12 03:08:00.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/7/20230112083316/NEC-FC : SUCCESS in 3h 31m 57s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/7/20230112083316/NEC-ISCSI : SUCCESS in 3h 34m 41s","accounts_in_message":[],"_revision_number":7},{"id":"cef577f7acce08f3b0b6d0977d997fec04e84437","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-01-19 11:36:18.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"17479a10bacbfe018fcaa5ac22471e9ad1ada14f","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-01-19 14:23:41.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bd89ce6c88e0eeaa1a85d9f289c1f4f0ca3854d2","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-01-20 15:39:51.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d3347036c95609db10d1bd77123eb044c1d13631","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-01-31 01:17:41.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"2bc0326947ca1d794c088060e60d4ae69badcddd","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-01-31 02:24:27.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"dc5b7d33d4da0b0fe58ec976a534e340164524ed","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-05-19 14:27:42.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1d9d2df2b9130e99ab04231853559e665cb985f6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-05-23 08:57:54.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a4bdf567cbf3831b9468627a1dcfc5c722f72776","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-05-28 05:04:20.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"0a21069c2fe8df8784812e163d8fbed4da8a8f2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-06-28 11:05:50.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":8},{"id":"86be75d5ea1eda68ed9c81cd496398e9faa02f5d","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-06-28 11:06:15.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"cd9fa75fcba541152f179bca7359858b98184a85","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-28 11:06:53.000000000","message":"Patch Set 8:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":8},{"id":"c8b336bd080215b682b5dc0ba92a44d757b2bb28","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-06-28 11:10:24.000000000","message":"Patch Set 8:\n\nBuild failed.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/00/848900/8/check/tyds_test/d69f113 : FAILURE in 4m 00s","accounts_in_message":[],"_revision_number":8},{"id":"2fb265ec82ee730ddcd3f9fcbf4dec81abb7aa69","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-06-28 11:48:46.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a7f713656b4678393400ba5d83cb251fd91837cb","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-28 13:10:59.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/ba4587a043f04e859e80414a78faf0b0\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7c4267d1b72e4e249db50d2480991477 : SUCCESS in 1h 14m 58s","accounts_in_message":[],"_revision_number":8},{"id":"5c75d45439271d39443d2865a6c565f1e33cf9f6","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-06-28 13:40:57.000000000","message":"Patch Set 8:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/00/848900/8/check/fujitsu-eternus-dx-iscsi/131c07f : FAILURE in 2h 08m 27s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/00/848900/8/check/fujitsu-eternus-dx-fc/fa7814b : FAILURE in 1h 48m 49s","accounts_in_message":[],"_revision_number":8},{"id":"d1e17510f103014825e0e8a986c0cedfad7dfdb8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-28 15:06:30.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/00/848900/8/check/iscsi-driver-nimble/a8aebbe/a8aebbe.html : FAILURE in 1h 10m 50s","accounts_in_message":[],"_revision_number":8},{"id":"d40e3358ee1c9bccddd5bc19897367b4f2df492e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-28 16:07:21.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-00-848900-8/ns5_nfs : FAILURE in 3h 56m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-00-848900-8/ns5_iscsi : FAILURE in 4h 19m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"57567f3aa92fd0ef1837a0478d149e982a5b0768","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-28 16:18:34.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/00/848900/8/check/DellEMC_PowerFlex/b86d428/DellEMC_PowerFlex/226 : SUCCESS in 1h 26m 19s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"5609013ee0f48c8cc7067a735a16f32011967e7d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-28 16:56:47.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/00/848900/8/check/DellEMC_Unity_FC/ef21280/DellEMC_Unity_FC/348 : FAILURE in 1h 43m 28s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/00/848900/8/check/DellEMC_Unity_iSCSI/cc305b5/DellEMC_Unity_iSCSI/347 : FAILURE in 1h 58m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"7563c1f8fdee03a9315fc5f431ddec956a97b403","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-28 19:00:08.000000000","message":"Patch Set 8:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/00/848900/8/fc : SUCCESS in 0h 49m 32s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/00/848900/8/iscsi : SUCCESS in 0h 44m 40s","accounts_in_message":[],"_revision_number":8},{"id":"e3cec3bb0ac67f4320ee293749c358dfdbec530a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-06-29 02:04:20.000000000","message":"Patch Set 8:\n\nBuild unstable. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1521/","accounts_in_message":[],"_revision_number":8},{"id":"6693c28cdb29413eb48577d3a409c3cb4c8efe22","author":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"date":"2023-06-29 15:59:18.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7d0ae29b8a85898985e93e6eaad4ab03a160d37b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-06-30 00:20:27.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/00/848900/8/check/DellEMC_PowerMAX_FC/22c326b/DellEMC_PowerMAX_FC/162 : SUCCESS in 7h 25m 10s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/00/848900/8/check/DellEMC_PowerMAX_iSCSI/7fa5f95/DellEMC_PowerMAX_iSCSI/156 : FAILURE in 8h 03m 35s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"84e67e679281b7e960b3b344ed7624e18d6cad21","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-30 13:26:57.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-00-848900-8/FC/ : SUCCESS in 2h 53m 2s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-00-848900-8/iSCSI/ : SUCCESS in 3h 31m 30s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"8335657d8dae5265a6209d398414baed2d37a6ee","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-07-03 10:27:35.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7cab8bdfe0c07a8a3bfef6b67a94f7ed5ab704c8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-04 14:12:20.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/8/20230704144930/NEC-ISCSI : FAILURE in 7h 48m 32s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/8/20230704144930/NEC-FC : FAILURE in 8h 22m 47s","accounts_in_message":[],"_revision_number":8},{"id":"f9110d7df87651b35525b8ddcad7bab57102e905","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-04 14:19:26.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/8/20230704165405/NEC-V-ISCSI : FAILURE in 6h 08m 07s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/8/20230704165405/NEC-V-FC : FAILURE in 6h 25m 18s","accounts_in_message":[],"_revision_number":8},{"id":"79ae853d842856b58b8095512b53418693447bbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-07-11 19:56:08.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":9},{"id":"308587f3d13a9e8e9e013a5dd2b33c281effcebf","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-11 19:58:00.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-00-848900-9/ns5_nfs : FAILURE in 1m 18s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-00-848900-9/ns5_iscsi : FAILURE in 1m 20s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"5e8c86f5da03dd0972ac30a06037a10a5234ebc8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-11 20:01:28.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f22a3478cb554ce6b301b41bfc666d6f : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/25b9537a5d4b4bfa83dff0c315e76b59 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"19704b533a1fd4e5f41e290ebe6a570c3050b45b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 20:21:11.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26bfa66dc8b24f1fa44eefcd4e5d8181\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f8298dac7ca4d78b26af2281f236d55 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/974ed81cc62645a59c8b5662ae94e0f6 : SUCCESS in 11m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6f5c4cb69b0d472fbf580b756003bc62 : SUCCESS in 23m 09s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"62185cab4cd096e336d86abf82c6948d19364262","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-11 21:13:01.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/16b9a919cb6947e38597c9cae7d670a6\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/40663d1ab80f48fa8969e325a7caf9e7 : FAILURE in 1h 13m 47s","accounts_in_message":[],"_revision_number":9},{"id":"3946cac652ef0492ddd155263d5df4dcc04e2c88","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-11 21:25:17.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/848900 : SUCCESS in 1h 27m 23s","accounts_in_message":[],"_revision_number":9},{"id":"ba92c2cfde180529c791f4ddd66453c72b60ce41","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-11 22:00:07.000000000","message":"Patch Set 9:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/00/848900/9/fc : SUCCESS in 0h 32m 18s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/00/848900/9/iscsi : SUCCESS in 0h 44m 27s","accounts_in_message":[],"_revision_number":9},{"id":"bb36fa1ef8d38fc79b6159671871c49ed95803d1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-11 22:06:10.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/00/848900/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/04cc81e/ : SUCCESS in 2h 08m 13s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/00/848900/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/9f1fbfd/ : SUCCESS in 1h 58m 58s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/00/848900/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/c259305/ : SUCCESS in 1h 49m 32s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/00/848900/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/2914d34/ : SUCCESS in 1h 52m 15s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"dcb95d19b64089ee09573dce14ca007cdabd6569","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-11 22:23:58.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-00-848900-9/FC/ : SUCCESS in 2h 27m 28s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-00-848900-9/iSCSI/ : SUCCESS in 2h 6m 8s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"48395b1bac45332ae1ddce6d74c5739d722526df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 22:27:42.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d017784c8bd4c84b6d128b93b911c17\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f4e2ed2af364ad586612ffb8fad1f32 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd4be4ecad34471996f1c3920546f849 : SUCCESS in 11m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d2d8e87b0f940ba83b8b19d44132914 : SUCCESS in 11m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f5ed7d6ab6064f089f1e6ad71213b519 : SUCCESS in 11m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51ce9ddf8a9248e8af58182c61e4003c : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/e121c11ffcab44588cfdc12e52c3ccb8 : SUCCESS in 1h 04m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/38f17e7dda134eb3a36d515d267bbf26 : SUCCESS in 1h 46m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6d9da93a46f45c1a30eba493f76f6ba : SUCCESS in 55m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae596112db01488784fa256eb8c2436d : SUCCESS in 6m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5878eb202a63408eac6fc2b4c0258d11 : SUCCESS in 16m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f52de7b1e5f14fada62dbd947eaefffd : SUCCESS in 5m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f10119c0697749f3a03f774ff9691901 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d2811d9863454ec18015f998d4399f01 : SUCCESS in 7m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/080639194ccd4845854bad2525bee660 : SUCCESS in 5m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/59f7a3d659ed454fa5c972eeaaf015de : SUCCESS in 35m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/deba06b0435f42d6a0a727b166a92c76 : SUCCESS in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a316843ccc3f4c84adfee19a8cd0daa1 : SUCCESS in 2h 26m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/49f49ab3954f4a7ca3e1f67f8e42d4b8 : FAILURE in 2h 00m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0485ebfa4f1744c589f22c991201eb88 : SUCCESS in 2h 06m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/502b9581a78a42d6926e6dcbd08966b2 : SUCCESS in 2h 01m 14s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4b88f69fb040457e87181eafba6092fb : SUCCESS in 1h 02m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9f2a96507d524259ab8fb9a877b49eb6 : SUCCESS in 1h 57m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d00bd3ce184b4d00baa3d8d32cee01af : SUCCESS in 54m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9b9138242620498d935de97af016d9a2 : SUCCESS in 1h 32m 52s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4aff25178a4f49ecb532dee1762a387c : POST_FAILURE in 2h 06m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9178732900624e5bab7e839caf4503a1 : FAILURE in 1h 49m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c191b0e1d9242f786f6d68dda4f1713 : SUCCESS in 1h 04m 05s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/dcb130b5455042b29ce3cba431758381 : SUCCESS in 9m 03s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"944e7c5498d1c6b58fab4a22ef4ac1ea23889108","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-12 00:48:50.000000000","message":"Patch Set 9:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1569/","accounts_in_message":[],"_revision_number":9},{"id":"4c5495d14b12b2c83fb2e013fb055a0d3cd5045f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-12 02:50:30.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/00/848900/9/check/DellEMC_Unity_FC/2b6f396/DellEMC_Unity_FC/409 : SUCCESS in 1h 42m 17s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/00/848900/9/check/DellEMC_Unity_iSCSI/991a28a/DellEMC_Unity_iSCSI/411 : SUCCESS in 2h 04m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"f369bfa39e13a190a17e237a66bf11156eb59d40","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-07-12 06:04:49.000000000","message":"Patch Set 9:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/00/848900/9/check/tyds_test/aff0ebb : SUCCESS in 3h 46m 55s","accounts_in_message":[],"_revision_number":9},{"id":"2dc07e138a7c9f1834d42e42b55ca2a33695f25c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-13 03:00:50.000000000","message":"Patch Set 9:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/9/20230713054250/NEC-V-FC : SUCCESS in 5h 54m 41s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/9/20230713054250/NEC-V-ISCSI : FAILURE in 6h 17m 57s","accounts_in_message":[],"_revision_number":9},{"id":"c664da11fa046f5356217fc4225b1a7e2c5f2340","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-13 08:45:24.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/9/20230713094445/NEC-ISCSI : SUCCESS in 3h 53m 41s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/848900/9/20230713094445/NEC-FC : SUCCESS in 8h 00m 36s","accounts_in_message":[],"_revision_number":9},{"id":"c4340ab82e03d630069b9c9bf0d4da39ebcd12a7","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-09-12 13:26:08.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9144e5c3c4b0c4926dabf8e77e19ef048fe6ca06","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-09-12 13:28:06.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/00/848900/9/check/fujitsu-eternus-dx-iscsi/ac6c711 : FAILURE in 1m 16s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/00/848900/9/check/fujitsu-eternus-dx-fc/1890b86 : FAILURE in 1m 20s","accounts_in_message":[],"_revision_number":9},{"id":"0fd18dd73f526f966a7e336a3b57ac1c5329f291","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-12 13:34:59.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/66414bacf03d4ada89ecaa3db969034a : FAILURE in 7m 06s","accounts_in_message":[],"_revision_number":9},{"id":"d1cbef8f6a98bb259790c2ffb0395eb9eeb2ca8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 13:59:41.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f058828f28ec40c88e4e32ec52ec45c9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/95fbd755f5604ec2bc72eced11f50463 : SUCCESS in 23m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/037451e11f534cbbb117c4260aa2abfe : SUCCESS in 28m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c30f74c8006435b8a83f184abf6b8a7 : SUCCESS in 30m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9680dbddecfa48febcfbed6b43e67822 : SUCCESS in 24m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f673b3d964f3c31906fdac03d63096cbba94a7cc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-09-12 14:56:25.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b1afde2d65184ef68a0861a8f4fef39c\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/aaf83d8d2f494af8ac97ccf5bd7e5402 : FAILURE in 1h 20m 07s","accounts_in_message":[],"_revision_number":9},{"id":"944eb847fe9a9c9824f027b7f6c1bdc667f59354","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-12 15:34:14.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/00/848900/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/b9da080/ : SUCCESS in 1h 37m 38s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/00/848900/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/3de0120/ : FAILURE in 2h 05m 57s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/00/848900/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/a0142c1/ : SUCCESS in 1h 45m 10s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/00/848900/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/0de7161/ : SUCCESS in 1h 43m 03s","accounts_in_message":[],"_revision_number":9},{"id":"cc826180b00b96591e27c1974729a5a5f49b73da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 15:46:08.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b009ec19b5e4ea3b53288e4251a9b85\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/360d7588a6c34928b088944dc5e12222 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cd5d5dedc89451ebe2adc2cb5a19802 : SUCCESS in 10m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/575cee22a046448c8d0558d08ebde6ea : SUCCESS in 10m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fea1406b721e4842825f844b10d79deb : SUCCESS in 11m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7fc9d89c942d4462911b4ef2db7cce29 : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c78a3767125471fbd00e4f89954c3b2 : SUCCESS in 14m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/09f246d5dff44d73a6027654453ef0e2 : SUCCESS in 1h 03m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/91792faa8c674e768e91684391a66867 : SUCCESS in 1h 19m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/afcc111c963c40e2a698ffe5daeaec9e : SUCCESS in 1h 01m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65bab5de24904a5f96d01fa33c57b1a4 : SUCCESS in 8m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/48a6af59ebac4ced8525725522f6d725 : SUCCESS in 16m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/11af8c6cc0bc43659d1fd7c8cf361990 : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ad6509942ec340e7a22e563075f7058e : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ec4baaa06a0b41ab9f8679c5d871699d : SUCCESS in 6m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6457235c92964939a6cc4e6ac610f99e : SUCCESS in 6m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dbe1798dd78341f1afefd33529715be4 : SUCCESS in 44m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d4ebfa263c04464e98ad1d174369bcf1 : SUCCESS in 6m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fb1930b757574b9889abb5f9d62731aa : SUCCESS in 1h 43m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/688febb7c92d489cb2a42b4a97e3238d : FAILURE in 2h 14m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c885b7fbf6564be391954ecd0d179bed : SUCCESS in 2h 06m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/91a0bd7b1e5a4b81b4ad399892ba5e20 : SUCCESS in 1h 46m 22s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/99a9da1bdf464c6c8433b93e58326e6c : SUCCESS in 1h 05m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/05631a1c1cae4d54ac4b8184666570c0 : FAILURE in 2h 06m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6b3cfd84c03d4def87cb050fb37caeae : SUCCESS in 48m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1415f2e275504d2fb08cef2895f785eb : SUCCESS in 1h 15m 26s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8948d90a1a364b70b923fcaedb5beba3 : SUCCESS in 1h 28m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/783a68226bd84f389ed87569458a50ed : SUCCESS in 1h 22m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9af894fa16b4784819d81f11571eac2 : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":9},{"id":"e41e5cb079359a2e26fc9a234e52354a22755481","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-12 16:10:17.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/00/848900/9/check/DellEMC_PowerFlex/45fe9d4/DellEMC_PowerFlex/595 : FAILURE in 23m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"41ce8479f0a0d7856a876ec14e66a76660b6807c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-12 16:59:05.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/00/848900/9/check/DellEMC_PowerStore_iSCSI/2550531/DellEMC_PowerStore_iSCSI/289 : SUCCESS in 53m 04s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/00/848900/9/check/DellEMC_PowerStore_FC/e8cf09d/DellEMC_PowerStore_FC/298 : SUCCESS in 1h 12m 39s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/00/848900/9/check/DellEMC_PowerStore_NFS/394ff0d/DellEMC_PowerStore_NFS/296 : SUCCESS in 46m 50s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/00/848900/9/check/DellEMC_PowerStore_NVMeTCP/7d32e5d/DellEMC_PowerStore_NVMeTCP/286 : SUCCESS in 1h 06m 44s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"1f043e45c9db8ec431c0d4702b113dc46f68026f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-12 17:17:15.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/848900/9/upstream-check/cinder-tempest-plugin-ontap-fcp/e78774e/ : SUCCESS in 1h 29m 53s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/848900/9/upstream-check/cinder-tempest-plugin-ontap-iscsi/9d6dcab/ : SUCCESS in 45m 04s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/848900/9/upstream-check/cinder-tempest-plugin-ontap-nfs/ce5d931/ : SUCCESS in 44m 15s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/848900/9/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/79e099d/ : SUCCESS in 43m 37s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"d314d89d23d57582efa8ac58933d4ba1bb064f33","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-12 18:41:50.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/00/848900/9/check/DellEMC_Unity_FC/220bb66/DellEMC_Unity_FC/782 : SUCCESS in 2h 04m 16s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/00/848900/9/check/DellEMC_Unity_iSCSI/856dd44/DellEMC_Unity_iSCSI/791 : SUCCESS in 2h 03m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"42ab27e4001ebcee076b4374349b329788b49734","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-09-14 02:25:43.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/848900-9/HPMSA-iSCSI/ : SUCCESS in 1:44:51\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/848900-9/HPMSA-FC/ : SUCCESS in 1:23:02","accounts_in_message":[],"_revision_number":9},{"id":"f8a636082ec6756e6c85d61a883cf1a119493621","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-09-14 13:04:59.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/848900-9/PVME-iSCSI/ : FAILURE in 2:03:56\n- PVME-FC https://os-logs.seagate.com/dell-ci/848900-9/PVME-FC/ : SUCCESS in 1:41:47","accounts_in_message":[],"_revision_number":9},{"id":"229879959b6269bff7c766d860585364dc5bed10","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-12-15 20:55:14.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1c8f8d2f7a227e1ed39c74d621717a676a5ae770","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 20:55:56.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"fe7624319dbe333bc20a14cd77d91ce65de98aaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 23:04:10.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aef0d014f7c64dd4b0b687d44eb02208\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67698bbbe04540f78a2d3618a98b1074 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e42093b5e2044299185191d22768e73 : SUCCESS in 10m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d843009467bf4704a7260db59c5bf658 : SUCCESS in 9m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdab0b1bfcb1427daacd35f3793fbe97 : SUCCESS in 13m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/119e3d8c74c24b1fa6305830ae497fb0 : SUCCESS in 56m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c2d7a2d8962a4aaa8a0cf2c98c6e8583 : SUCCESS in 44m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3e49172306154ceba2dfa828cdb65469 : SUCCESS in 1h 16m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7aae6e2bc614419ba387e394bb620379 : SUCCESS in 54m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce410a60f21d487d8fa02fa760629f6b : SUCCESS in 8m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f35b2275f0e242feb0e8d37b72df8023 : SUCCESS in 54m 32s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8b4503d50fe24d759b4971f43ac2eded : SUCCESS in 1h 45m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc1665392f5147f593a6a1371b272edf : SUCCESS in 43m 17s","accounts_in_message":[],"_revision_number":9},{"id":"8ac707d3aeba5dab64a1dedf2389548467c43dcd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 23:04:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"c54b4b45ebf46f69d7a2732d4742804777e30fc0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 23:05:40.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15301b92115f4013b6c784b002167337\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/064bcb7dd7a8410c92571e16b14660ee : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e174e61ddca7434d83baea2817a5b565 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"411852892d803b606110d0956a59764925c16ec6","revisions":{"fb55152dd7a78dfaeaf9033f818250d1faea921c":{"kind":"REWORK","_number":1,"created":"2022-07-06 19:31:08.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/00/848900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/848900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/848900/1"}}},"commit":{"parents":[{"commit":"7b3d71bb02a2e233ef063609c78a746b5424c632","subject":"Deadlock prevention support in syncrhonize","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b3d71bb02a2e233ef063609c78a746b5424c632"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-07-06 18:51:34.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-07-06 19:26:18.000000000","tz":120},"subject":"PowerMax: Fix deadlock moving SGs","message":"PowerMax: Fix deadlock moving SGs\n\nThere\u0027s a potential deadlock scenario in PowerMax\u0027s masking.py\nmove_volume_between_storage_groups\" method.\n\nThe method uses 2 locks, one for the source Storage Group and another\nfor the destination Storage Group, and it could happen that if 2\nrequests going in opposite directions are received simultaneously their\nfirst lock acquisition interleaves resulting in a deadlock situation.\n\n    @coordination.synchronized(\n        \"emc-sg-{source_storagegroup_name}-{serial_number}\")\n    @coordination.synchronized(\n        \"emc-sg-{target_storagegroup_name}-{serial_number}\")\n    def move_volume_between_storage_groups(\n            self, serial_number, device_id, source_storagegroup_name,\n            target_storagegroup_name, extra_specs, force\u003dFalse,\n            parent_sg\u003dNone):\n\nThe scenario would be like this:\n\n- User requests an instance migration from A to B\n- User requests an instance migration from B to A\n- Driver acquires the first lock for A-to-B for example something like\n  cinder-emc-sg-SGA-###\n- Driver acquires the first lock for B-to-A for example something like\n  cinder-emc-sgSGB-###\n\nThe deadlock happens because A-to-B waits forever for the lock held by\nthe B-to-A operation, which in turn cannot proceed because it’s waiting\nfor lock help by A-to-B.\n\nThis patch fixes it using the new coordination.synchronized\nfunctionality that ensures that a series of locks are always acquired in\nthe same order, preventing deadlocks.\n\nChange-Id: I7eda4645575cfaedcf45d73ab3a215976d3fac3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fb55152dd7a78dfaeaf9033f818250d1faea921c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fb55152dd7a78dfaeaf9033f818250d1faea921c"}]},"branch":"refs/heads/master"},"52682e041c86a7b9f4b0af075799aafc44a1bd17":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-06 19:34:07.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/00/848900/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/848900/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/848900/2"}}},"commit":{"parents":[{"commit":"7b3d71bb02a2e233ef063609c78a746b5424c632","subject":"Deadlock prevention support in syncrhonize","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b3d71bb02a2e233ef063609c78a746b5424c632"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-07-06 18:51:34.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-07-06 19:33:50.000000000","tz":120},"subject":"PowerMax: Fix deadlock moving SGs","message":"PowerMax: Fix deadlock moving SGs\n\nThere\u0027s a potential deadlock scenario in PowerMax\u0027s masking.py\nmove_volume_between_storage_groups\" method.\n\nThe method uses 2 locks, one for the source Storage Group and another\nfor the destination Storage Group, and it could happen that if 2\nrequests going in opposite directions are received simultaneously their\nfirst lock acquisition interleaves resulting in a deadlock situation.\n\n    @coordination.synchronized(\n        \"emc-sg-{source_storagegroup_name}-{serial_number}\")\n    @coordination.synchronized(\n        \"emc-sg-{target_storagegroup_name}-{serial_number}\")\n    def move_volume_between_storage_groups(\n            self, serial_number, device_id, source_storagegroup_name,\n            target_storagegroup_name, extra_specs, force\u003dFalse,\n            parent_sg\u003dNone):\n\nThe scenario would be like this:\n\n- User requests an instance migration from A to B\n- User requests an instance migration from B to A\n- Driver acquires the first lock for A-to-B for example something like\n  cinder-emc-sg-SGA-###\n- Driver acquires the first lock for B-to-A for example something like\n  cinder-emc-sgSGB-###\n\nThe deadlock happens because A-to-B waits forever for the lock held by\nthe B-to-A operation, which in turn cannot proceed because it’s waiting\nfor lock help by A-to-B.\n\nThis patch fixes it using the new coordination.synchronized\nfunctionality that ensures that a series of locks are always acquired in\nthe same order, preventing deadlocks.\n\nCloses-Bug: #1980870\nChange-Id: I7eda4645575cfaedcf45d73ab3a215976d3fac3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/52682e041c86a7b9f4b0af075799aafc44a1bd17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/52682e041c86a7b9f4b0af075799aafc44a1bd17"}]},"branch":"refs/heads/master"},"5a8bd158c47eedc4bc844904d759120d3af10803":{"kind":"REWORK","_number":3,"created":"2022-07-12 09:20:21.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/00/848900/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/848900/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/848900/3"}}},"commit":{"parents":[{"commit":"7b3d71bb02a2e233ef063609c78a746b5424c632","subject":"Deadlock prevention support in syncrhonize","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b3d71bb02a2e233ef063609c78a746b5424c632"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-07-06 18:51:34.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-07-12 09:20:08.000000000","tz":120},"subject":"PowerMax: Fix deadlock moving SGs","message":"PowerMax: Fix deadlock moving SGs\n\nThere\u0027s a potential deadlock scenario in PowerMax\u0027s masking.py\nmove_volume_between_storage_groups\" method.\n\nThe method uses 2 locks, one for the source Storage Group and another\nfor the destination Storage Group, and it could happen that if 2\nrequests going in opposite directions are received simultaneously their\nfirst lock acquisition interleaves resulting in a deadlock situation.\n\n    @coordination.synchronized(\n        \"emc-sg-{source_storagegroup_name}-{serial_number}\")\n    @coordination.synchronized(\n        \"emc-sg-{target_storagegroup_name}-{serial_number}\")\n    def move_volume_between_storage_groups(\n            self, serial_number, device_id, source_storagegroup_name,\n            target_storagegroup_name, extra_specs, force\u003dFalse,\n            parent_sg\u003dNone):\n\nThe scenario would be like this:\n\n- User requests an instance migration from A to B\n- User requests an instance migration from B to A\n- Driver acquires the first lock for A-to-B for example something like\n  cinder-emc-sg-SGA-###\n- Driver acquires the first lock for B-to-A for example something like\n  cinder-emc-sgSGB-###\n\nThe deadlock happens because A-to-B waits forever for the lock held by\nthe B-to-A operation, which in turn cannot proceed because it’s waiting\nfor lock help by A-to-B.\n\nThis patch fixes it using the new coordination.synchronized\nfunctionality that ensures that a series of locks are always acquired in\nthe same order, preventing deadlocks.\n\nCloses-Bug: #1980870\nChange-Id: I7eda4645575cfaedcf45d73ab3a215976d3fac3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5a8bd158c47eedc4bc844904d759120d3af10803"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5a8bd158c47eedc4bc844904d759120d3af10803"}]},"branch":"refs/heads/master"},"4de4ecf8fdcf26c1581b0536950e800a81737cb0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-07-22 10:30:04.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/00/848900/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/848900/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/848900/4"}}},"commit":{"parents":[{"commit":"aa0a49cd8b665a4b211b0135008f3847f2361e1d","subject":"Deadlock prevention support in synchronize","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aa0a49cd8b665a4b211b0135008f3847f2361e1d"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-07-06 18:51:34.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-07-22 10:29:55.000000000","tz":120},"subject":"PowerMax: Fix deadlock moving SGs","message":"PowerMax: Fix deadlock moving SGs\n\nThere\u0027s a potential deadlock scenario in PowerMax\u0027s masking.py\nmove_volume_between_storage_groups\" method.\n\nThe method uses 2 locks, one for the source Storage Group and another\nfor the destination Storage Group, and it could happen that if 2\nrequests going in opposite directions are received simultaneously their\nfirst lock acquisition interleaves resulting in a deadlock situation.\n\n    @coordination.synchronized(\n        \"emc-sg-{source_storagegroup_name}-{serial_number}\")\n    @coordination.synchronized(\n        \"emc-sg-{target_storagegroup_name}-{serial_number}\")\n    def move_volume_between_storage_groups(\n            self, serial_number, device_id, source_storagegroup_name,\n            target_storagegroup_name, extra_specs, force\u003dFalse,\n            parent_sg\u003dNone):\n\nThe scenario would be like this:\n\n- User requests an instance migration from A to B\n- User requests an instance migration from B to A\n- Driver acquires the first lock for A-to-B for example something like\n  cinder-emc-sg-SGA-###\n- Driver acquires the first lock for B-to-A for example something like\n  cinder-emc-sgSGB-###\n\nThe deadlock happens because A-to-B waits forever for the lock held by\nthe B-to-A operation, which in turn cannot proceed because it’s waiting\nfor lock help by A-to-B.\n\nThis patch fixes it using the new coordination.synchronized\nfunctionality that ensures that a series of locks are always acquired in\nthe same order, preventing deadlocks.\n\nCloses-Bug: #1980870\nChange-Id: I7eda4645575cfaedcf45d73ab3a215976d3fac3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4de4ecf8fdcf26c1581b0536950e800a81737cb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4de4ecf8fdcf26c1581b0536950e800a81737cb0"}]},"branch":"refs/heads/master"},"7ccc896281122df00a0fb6d5048cdeba360b512e":{"kind":"REWORK","_number":5,"created":"2022-08-24 09:22:47.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/00/848900/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/848900/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/848900/5"}}},"commit":{"parents":[{"commit":"f21200dee1791ae40085657e44a578cebbd1463c","subject":"Deadlock prevention support in synchronize","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f21200dee1791ae40085657e44a578cebbd1463c"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-07-06 18:51:34.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-24 09:21:16.000000000","tz":120},"subject":"PowerMax: Fix deadlock moving SGs","message":"PowerMax: Fix deadlock moving SGs\n\nThere\u0027s a potential deadlock scenario in PowerMax\u0027s masking.py\n\"do_move_volume_between_storage_groups\" method.\n\nThe method uses 2 locks, one for the source Storage Group and another\nfor the destination Storage Group, and it could happen that if 2\nrequests going in opposite directions are received simultaneously their\nfirst lock acquisition interleaves resulting in a deadlock situation.\n\n    @coordination.synchronized(\n        \"emc-sg-{source_storagegroup_name}-{serial_number}\")\n    @coordination.synchronized(\n        \"emc-sg-{target_storagegroup_name}-{serial_number}\")\n    def do_move_volume_between_storage_groups(\n        serial_number, source_storage_group_name,\n        target_storage_group_name):\n\nThe scenario would be like this:\n\n- User requests an instance migration from A to B\n- User requests an instance migration from B to A\n- Driver acquires the first lock for A-to-B for example something like\n  cinder-emc-sg-SGA-###\n- Driver acquires the first lock for B-to-A for example something like\n  cinder-emc-sgSGB-###\n\nThe deadlock happens because A-to-B waits forever for the lock held by\nthe B-to-A operation, which in turn cannot proceed because it’s waiting\nfor lock help by A-to-B.\n\nThis patch fixes it using the new coordination.synchronized\nfunctionality that ensures that a series of locks are always acquired in\nthe same order, preventing deadlocks.\n\nCloses-Bug: #1980870\nChange-Id: I7eda4645575cfaedcf45d73ab3a215976d3fac3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ccc896281122df00a0fb6d5048cdeba360b512e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ccc896281122df00a0fb6d5048cdeba360b512e"}]},"branch":"refs/heads/master"},"4e6cbf1096f25e2c50178a7d2e1082d074b29111":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-08-24 13:49:49.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/00/848900/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/848900/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/848900/6"}}},"commit":{"parents":[{"commit":"dd084a15dece6eb24fc6ea450edc1ee8b6c245c9","subject":"Deadlock prevention support in synchronize","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dd084a15dece6eb24fc6ea450edc1ee8b6c245c9"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-07-06 18:51:34.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-24 13:49:40.000000000","tz":120},"subject":"PowerMax: Fix deadlock moving SGs","message":"PowerMax: Fix deadlock moving SGs\n\nThere\u0027s a potential deadlock scenario in PowerMax\u0027s masking.py\n\"do_move_volume_between_storage_groups\" method.\n\nThe method uses 2 locks, one for the source Storage Group and another\nfor the destination Storage Group, and it could happen that if 2\nrequests going in opposite directions are received simultaneously their\nfirst lock acquisition interleaves resulting in a deadlock situation.\n\n    @coordination.synchronized(\n        \"emc-sg-{source_storagegroup_name}-{serial_number}\")\n    @coordination.synchronized(\n        \"emc-sg-{target_storagegroup_name}-{serial_number}\")\n    def do_move_volume_between_storage_groups(\n        serial_number, source_storage_group_name,\n        target_storage_group_name):\n\nThe scenario would be like this:\n\n- User requests an instance migration from A to B\n- User requests an instance migration from B to A\n- Driver acquires the first lock for A-to-B for example something like\n  cinder-emc-sg-SGA-###\n- Driver acquires the first lock for B-to-A for example something like\n  cinder-emc-sgSGB-###\n\nThe deadlock happens because A-to-B waits forever for the lock held by\nthe B-to-A operation, which in turn cannot proceed because it’s waiting\nfor lock help by A-to-B.\n\nThis patch fixes it using the new coordination.synchronized\nfunctionality that ensures that a series of locks are always acquired in\nthe same order, preventing deadlocks.\n\nCloses-Bug: #1980870\nChange-Id: I7eda4645575cfaedcf45d73ab3a215976d3fac3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e6cbf1096f25e2c50178a7d2e1082d074b29111"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e6cbf1096f25e2c50178a7d2e1082d074b29111"}]},"branch":"refs/heads/master"},"1852eb0f412a8592f434fd1a3f31a0b0cba9cce1":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-01-10 08:03:41.000000000","uploader":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"ref":"refs/changes/00/848900/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/848900/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/848900/7"}}},"commit":{"parents":[{"commit":"3557fcd7c215408a5241c3b5df5a8c09c9138039","subject":"Deadlock prevention support in synchronize","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3557fcd7c215408a5241c3b5df5a8c09c9138039"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-07-06 18:51:34.000000000","tz":120},"committer":{"name":"Jean-Pierre Roquesalane","email":"jeanpierre.roquesalane@dell.com","date":"2023-01-10 08:03:07.000000000","tz":60},"subject":"PowerMax: Fix deadlock moving SGs","message":"PowerMax: Fix deadlock moving SGs\n\nThere\u0027s a potential deadlock scenario in PowerMax\u0027s masking.py\n\"do_move_volume_between_storage_groups\" method.\n\nThe method uses 2 locks, one for the source Storage Group and another\nfor the destination Storage Group, and it could happen that if 2\nrequests going in opposite directions are received simultaneously their\nfirst lock acquisition interleaves resulting in a deadlock situation.\n\n    @coordination.synchronized(\n        \"emc-sg-{source_storagegroup_name}-{serial_number}\")\n    @coordination.synchronized(\n        \"emc-sg-{target_storagegroup_name}-{serial_number}\")\n    def do_move_volume_between_storage_groups(\n        serial_number, source_storage_group_name,\n        target_storage_group_name):\n\nThe scenario would be like this:\n\n- User requests an instance migration from A to B\n- User requests an instance migration from B to A\n- Driver acquires the first lock for A-to-B for example something like\n  cinder-emc-sg-SGA-###\n- Driver acquires the first lock for B-to-A for example something like\n  cinder-emc-sgSGB-###\n\nThe deadlock happens because A-to-B waits forever for the lock held by\nthe B-to-A operation, which in turn cannot proceed because it’s waiting\nfor lock help by A-to-B.\n\nThis patch fixes it using the new coordination.synchronized\nfunctionality that ensures that a series of locks are always acquired in\nthe same order, preventing deadlocks.\n\nCloses-Bug: #1980870\nChange-Id: I7eda4645575cfaedcf45d73ab3a215976d3fac3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1852eb0f412a8592f434fd1a3f31a0b0cba9cce1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1852eb0f412a8592f434fd1a3f31a0b0cba9cce1"}]},"branch":"refs/heads/master"},"78407657ea9bf72aeab6b63958a2976ef25292f5":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2023-06-28 11:05:50.000000000","uploader":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"ref":"refs/changes/00/848900/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/848900/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/848900/8"}}},"commit":{"parents":[{"commit":"3557fcd7c215408a5241c3b5df5a8c09c9138039","subject":"Deadlock prevention support in synchronize","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3557fcd7c215408a5241c3b5df5a8c09c9138039"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-07-06 18:51:34.000000000","tz":120},"committer":{"name":"Jean Pierre Roquesalane","email":"jeanpierre.roquesalane@dell.com","date":"2023-06-28 11:05:50.000000000","tz":0},"subject":"PowerMax: Fix deadlock moving SGs","message":"PowerMax: Fix deadlock moving SGs\n\nThere\u0027s a potential deadlock scenario in PowerMax\u0027s masking.py\n\"do_move_volume_between_storage_groups\" method.\n\nThe method uses 2 locks, one for the source Storage Group and another\nfor the destination Storage Group, and it could happen that if 2\nrequests going in opposite directions are received simultaneously their\nfirst lock acquisition interleaves resulting in a deadlock situation.\n\n    @coordination.synchronized(\n        \"emc-sg-{source_storagegroup_name}-{serial_number}\")\n    @coordination.synchronized(\n        \"emc-sg-{target_storagegroup_name}-{serial_number}\")\n    def do_move_volume_between_storage_groups(\n        serial_number, source_storage_group_name,\n        target_storage_group_name):\n\nThe scenario would be like this:\n\n- User requests an instance migration from A to B\n- User requests an instance migration from B to A\n- Driver acquires the first lock for A-to-B for example something like\n  cinder-emc-sg-SGA-###\n- Driver acquires the first lock for B-to-A for example something like\n  cinder-emc-sgSGB-###\n\nThe deadlock happens because A-to-B waits forever for the lock held by\nthe B-to-A operation, which in turn cannot proceed because it’s waiting\nfor lock held by A-to-B.\n\nThis patch fixes it using the new coordination.synchronized\nfunctionality that ensures that a series of locks are always acquired in\nthe same order, preventing deadlocks.\n\nCloses-Bug: #1980870\nChange-Id: I7eda4645575cfaedcf45d73ab3a215976d3fac3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78407657ea9bf72aeab6b63958a2976ef25292f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78407657ea9bf72aeab6b63958a2976ef25292f5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"411852892d803b606110d0956a59764925c16ec6":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-07-11 19:56:08.000000000","uploader":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"ref":"refs/changes/00/848900/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/848900/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/848900/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/848900/9"}}},"commit":{"parents":[{"commit":"c1a18fcdf963a512b09b9007f6cc0803415e166e","subject":"Merge \"Ceph: Fix restore backups to diff backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1a18fcdf963a512b09b9007f6cc0803415e166e"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-07-06 18:51:34.000000000","tz":120},"committer":{"name":"Jean-Pierre Roquesalane","email":"jeanpierre.roquesalane@dell.com","date":"2023-07-11 19:56:04.000000000","tz":120},"subject":"PowerMax: Fix deadlock moving SGs","message":"PowerMax: Fix deadlock moving SGs\n\nThere\u0027s a potential deadlock scenario in PowerMax\u0027s masking.py\n\"do_move_volume_between_storage_groups\" method.\n\nThe method uses 2 locks, one for the source Storage Group and another\nfor the destination Storage Group, and it could happen that if 2\nrequests going in opposite directions are received simultaneously their\nfirst lock acquisition interleaves resulting in a deadlock situation.\n\n    @coordination.synchronized(\n        \"emc-sg-{source_storagegroup_name}-{serial_number}\")\n    @coordination.synchronized(\n        \"emc-sg-{target_storagegroup_name}-{serial_number}\")\n    def do_move_volume_between_storage_groups(\n        serial_number, source_storage_group_name,\n        target_storage_group_name):\n\nThe scenario would be like this:\n\n- User requests an instance migration from A to B\n- User requests an instance migration from B to A\n- Driver acquires the first lock for A-to-B for example something like\n  cinder-emc-sg-SGA-###\n- Driver acquires the first lock for B-to-A for example something like\n  cinder-emc-sgSGB-###\n\nThe deadlock happens because A-to-B waits forever for the lock held by\nthe B-to-A operation, which in turn cannot proceed because it’s waiting\nfor lock held by A-to-B.\n\nThis patch fixes it using the new coordination.synchronized\nfunctionality that ensures that a series of locks are always acquired in\nthe same order, preventing deadlocks.\n\nCloses-Bug: #1980870\nChange-Id: I7eda4645575cfaedcf45d73ab3a215976d3fac3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/411852892d803b606110d0956a59764925c16ec6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/411852892d803b606110d0956a59764925c16ec6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
