)]}'
{"id":"openstack%2Fcinder~849626","triplet_id":"openstack%2Fcinder~master~I0b389f4f53d3b34eddc23a7aebc85a5b42f3a09f","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-07-15 18:24:36.000000000","reason":"Change was submitted"},"33502":{"account":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"last_update":"2022-07-15 18:24:36.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-07-15 14:28:19.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I0b389f4f53d3b34eddc23a7aebc85a5b42f3a09f","subject":"PowerStore driver - Request data validation fix","status":"MERGED","created":"2022-07-13 08:08:41.000000000","updated":"2022-07-18 01:20:31.000000000","submitted":"2022-07-15 18:24:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"849626","meta_rev_id":"1e76096e270a10a608c91bb9df0b810477d0210c","_number":849626,"virtual_id_number":849626,"owner":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 18:24:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"all":[{"value":0,"date":"2022-07-15 11:11:24.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-15 14:17:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":2,"date":"2022-07-15 14:28:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":1,"date":"2022-07-15 14:28:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-13 08:09:14.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-13 08:12:02.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-07-13 08:38:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-13 09:06:02.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-13 09:14:10.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-07-13 09:19:53.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-07-13 09:31:28.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-13 09:52:13.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-13 10:00:28.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-07-13 10:32:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-13 10:40:40.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-07-13 11:04:51.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-13 11:15:03.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-07-13 11:37:17.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-13 12:16:18.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-13 13:27:38.000000000","updated_by":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-07-13 13:27:38.000000000","updated_by":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-07-13 13:44:34.000000000","updated_by":{"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},"reviewer":{"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},"state":"REVIEWER"},{"updated":"2022-07-13 16:24:18.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-07-13 23:45:24.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-07-14 09:12:16.000000000","updated_by":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"reviewer":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"state":"CC"},{"updated":"2022-07-14 10:05:36.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-07-15 14:17:20.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2022-07-18 01:19:30.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"71341d7ca318cc97ce0607acadd2cb0972b415c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"date":"2022-07-13 08:08:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec155f731cb7eaa39ea0c0e78db85f8a54b64423","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-13 08:09:14.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"5a39ddab67ecdab6360856422ca23adcd7506da2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-13 08:12:02.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1657699732690 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ade347d5b8ba4151edd795f3b3bbfb18f175c6dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"date":"2022-07-13 08:12:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f28e0a19ffcb5ef8fa498a1cfa97110adbeaf5f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-13 08:13:31.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"393d8ca445ffc385890cc2b82f36a6b96422241c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"date":"2022-07-13 08:14:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"65056d281faf8063159679ba04baef06e8b47f11","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-13 08:15:20.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"3322ec3908f6b28bd7aea7de281efa9f441a5ef8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-13 08:16:15.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1657699987695 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"d1e98c25aa7fc5f88b2ce142e4375dc15c9e847f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-13 08:35:57.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"805f34fb2f60ee2a97d624424c0bed54237f71ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 08:38:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98d710bcff9446c0acc78f7ef88934e2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eec5136d336d4e8a8221ee5abe190075 : SUCCESS in 13m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/efcc0e618acf487698e2e0e914c5e08d : SUCCESS in 21m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6598bece596d7eac73b033aa44fb488ec1afa8ad","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-07-13 09:06:02.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/26/849626/3/ : RETRY_LIMIT in 23s","accounts_in_message":[],"_revision_number":3},{"id":"7e80161186f5eaf51cf2f35709289e14e999882d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-13 09:14:10.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/26/849626/3/check/dsvm-tempest-cisco-zonemanager-job/77fc33f/ : FAILURE in 52m 23s","accounts_in_message":[],"_revision_number":3},{"id":"3e6e6b9f75aeaaa53a607adbf5aac51fb923ddbf","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-13 09:19:53.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/849626/3/iscsi/ : SUCCESS in 1h 04m 38s\n- cinder-smb http://cloudbase-ci.com/cinder/849626/3/smb/ : SUCCESS in 54m 02s","accounts_in_message":[],"_revision_number":3},{"id":"68feb5fe374d57fb4ff2c6a5bbebb5c93ecce51a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-07-13 09:31:28.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5122bde82fd345219fd48bc5c6cc6763 : FAILURE in 1h 15m 09s","accounts_in_message":[],"_revision_number":3},{"id":"040afe61b08b46a0dc9b9d0fa40fc7c711339755","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-13 09:52:13.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/849626 : FAILURE in 1h 34m 23s","accounts_in_message":[],"_revision_number":3},{"id":"297822680a77ee4b577c1986840dc0fb1df811e1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-13 10:00:28.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/26/849626/3/check/iscsi-driver-nimble/b9233c5/b9233c5.html : FAILURE in 1h 43m 16s","accounts_in_message":[],"_revision_number":3},{"id":"9efd643074f195f063c8c308d873ee4c6cc78e48","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-13 10:30:59.000000000","message":"Patch Set 3:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/26/849626/3/check/fujitsu-eternus-dx-iscsi/20aa5e4 : SUCCESS in 1h 48m 54s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/26/849626/3/check/fujitsu-eternus-dx-fc/3975a11 : SUCCESS in 1h 50m 32s","accounts_in_message":[],"_revision_number":3},{"id":"39a441a8797bd476a2864ec01e07c84712744c63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 10:32:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eadf3361ea4e4488941160bebf84f43a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05e7fa6568584e79af945827ec792c82 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d0b5406e21d4bd8a97cba6bcd829793 : SUCCESS in 11m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5607e23cef04a3ca8ed0b217a5703ed : SUCCESS in 10m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7362a2718c99418c89a158c84d28983b : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/438e9b948cc24ca2b34034b54aeb71a0 : SUCCESS in 10m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/7e5ea9c910894aa592a61dcea8e8b15b : SUCCESS in 54m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a87371417292435a90036d4875c92aed : SUCCESS in 1h 34m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec897de5e3b645a893b26fe029bd6c2a : SUCCESS in 47m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d04baded650403a95014bb804d4945a : SUCCESS in 9m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/626a1ca5ce1244f787ff48c184eff587 : SUCCESS in 14m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ac58fd1a30004ad68d246ac39654792f : SUCCESS in 6m 00s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0e37e2f8bb1941e9bf199956ec4ecefd : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ed63003ce13042cd82e216b3b2c4793e : SUCCESS in 6m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e0262d7cadf947788fe3363b4d4fd8c0 : SUCCESS in 5m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8261e40d57c0497c830b22048a324fde : SUCCESS in 49m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3a5790c9408a4ad4bab2e9712407beb2 : FAILURE in 5m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8b60d9b70cc24bc0b51f2456bfd1a4a9 : SUCCESS in 1h 30m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/751420c221534232810f9aad32f277c0 : FAILURE in 1h 25m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6005712dcd454ce8a7bb138a238898b5 : SUCCESS in 1h 56m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/38556a58e87f47d5986d5621c927c343 : SUCCESS in 1h 05m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/171a3934a9344592b0f4da5b7001192c : SUCCESS in 1h 16m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ed3098a2d6b346d88d06956a936ff7e8 : SUCCESS in 38m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/68547fa36db34e398d84469acd0fa421 : SUCCESS in 1h 16m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/88977bcad2704a218b96be887f2632cf : SUCCESS in 2h 12m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/885f983acd224b86831007dc0d3fa142 : SUCCESS in 51m 40s","accounts_in_message":[],"_revision_number":3},{"id":"a75a020fcbbe185ecfcfb86430f8e61a6742e758","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-13 10:40:40.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp cinder-tempest-plugin-ontap-fcp : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/849626/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/a6cfd0c/ : RETRY_LIMIT in 7m 05s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/849626/3/upstream-check/cinder-tempest-plugin-ontap-nfs/bf786d3/ : RETRY_LIMIT in 7m 06s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"b975fea4c3c62a3d2d11f05b0bcf4aa64be57e1a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-13 11:04:51.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/26/849626/3/check/tempest-dsvm-full-focal-py3/c896507/ : SUCCESS in 1h 15m 47s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/26/849626/3/check/devstack-plugin-ceph-dsvm-focal-py3/3b9b645/ : FAILURE in 2h 47m 12s","accounts_in_message":[],"_revision_number":3},{"id":"b25b77764b798a6e7a13006bcbd537e05d1e4f96","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-13 11:15:03.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/26/849626/3/check/DellEMC_PowerStore_NFS/c7ca7a3/DellEMC_PowerStore_NFS/118 : FAILURE in 42m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"b49125ee9dbad51f151b2362e4be95a49d18757f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-13 11:37:17.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-26-849626-1/ns5_nfs : SUCCESS in 3h 23m 20s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-26-849626-1/ns5_iscsi : SUCCESS in 3h 28m 4s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"2d53548491a19abc947ac7eb910363e27d9ee8f4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-13 11:58:23.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-26-849626-2/ns5_nfs : SUCCESS in 3h 21m 32s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-26-849626-2/ns5_iscsi : SUCCESS in 3h 26m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"9d25e9d033e3e1cddea74c69344ed864ea70d296","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-13 12:16:18.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/26/849626/3/check/DellEMC_PowerFlex/ebe881d/DellEMC_PowerFlex/983 : SUCCESS in 1h 43m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"bccd734d783b2f3462cb4568f3adda6ce1d66571","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-13 12:24:41.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-26-849626-3/ns5_nfs : SUCCESS in 3h 28m 30s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-26-849626-3/ns5_iscsi : SUCCESS in 3h 30m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"90d7a25f7d4252227a6fd8787b98898eff71c876","author":{"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},"date":"2022-07-13 13:44:34.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2b1b18154e5f75e9c99b5a2b0639141c504d7cec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"date":"2022-07-13 14:38:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e17cc6924014e8f3d1415a1d5f9cd41ad1625998","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 15:10:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/646c14bd7a7046b8a3543b7dcd3e3b9c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b15814a50f844b6e8f54c2da00f08cd9 : SUCCESS in 26m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5d345e54784646a0bfdd8c11cbcffee1 : SUCCESS in 22m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fdb1f0094e8934e09bc4d4d5c4caf940d05e9850","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-13 15:23:45.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1657723148347 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"304ff009043b2c7411dfb1021e17d9405663493f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-13 15:45:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/849626/4/iscsi/ : SUCCESS in 1h 05m 49s\n- cinder-smb http://cloudbase-ci.com/cinder/849626/4/smb/ : SUCCESS in 1h 02m 34s","accounts_in_message":[],"_revision_number":4},{"id":"dc7042aa3456873c78209f53b3a1c900f8e528c6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-07-13 16:01:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ac44e2d8612c478cabda5b0254a05b7d : SUCCESS in 37m 54s","accounts_in_message":[],"_revision_number":4},{"id":"0d0723ec9517d68b4b28dcfaf5dfd513b1ca42bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 16:23:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/31198d380d7e4c1785016f00f8bfb3d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b27b0bef4aa441149c5487fa9ec96d87 : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c4c18777c8e4df0ac007941c5807759 : SUCCESS in 14m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d71c00fce0e24b879fe2fe87f7fcf809 : SUCCESS in 13m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de0c8bdd6e1f41539a5c1cdf14f41d84 : SUCCESS in 14m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c39bfd588e834b8bb25c7fcc71635d9a : SUCCESS in 13m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/f84f06f89dac43faa2c7c306136cee6d : SUCCESS in 1h 02m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/38db50f3bbe24071aed8a33dfc55e4d5 : SUCCESS in 1h 25m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/044f173f7b0445feac42cea429e8f52a : SUCCESS in 53m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae25701fcd29462cb8b52c02d9201190 : SUCCESS in 9m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4eff65402f6246279ca789efafbdfc11 : SUCCESS in 16m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/580f98b0a6884b949e048aab53174dbf : SUCCESS in 6m 09s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5e032df3076b4030bb738b0c5dc7939d : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c6803fc072d94f95b2c0b95c08afcf66 : SUCCESS in 7m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d035aa33f42d43f68c503615923925eb : SUCCESS in 7m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2fe36d4b84a24765899e2c06e903b389 : SUCCESS in 37m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8d00fc2d81f148909116d9babab2419f : FAILURE in 7m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/374f4035aeeb4ea2a59664f23793ac60 : SUCCESS in 1h 14m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1cbb58540428480eb311103157396de3 : FAILURE in 1h 28m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c832a5937348450a91e54f9c666b8f4e : SUCCESS in 1h 31m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ffbd6e1a12174c77af097f6faf861b3c : SUCCESS in 1h 01m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1f64a0f2399a4bdb817f1dbfb3850025 : SUCCESS in 1h 16m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/08b385f512cb48eabbc90cb39ca671c4 : SUCCESS in 43m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/19eae1fa54f747869e8a7012f20ad7a2 : SUCCESS in 1h 28m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a1e237b53be846f489fccddcd752b27a : RETRY_LIMIT in 6m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7bb5480fe8e4b719a40d7fdbba3e9c8 : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":4},{"id":"119be8e8abcfebdd744e0f636960cebe15461aa1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-13 16:24:18.000000000","message":"Patch Set 4:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/6","accounts_in_message":[],"_revision_number":4},{"id":"38b96ec86cd0197122145c38d10dd2da4cfc4614","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-13 16:29:04.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/849626 : FAILURE in 1h 44m 41s","accounts_in_message":[],"_revision_number":4},{"id":"0b73970ba5705b5fa07cb89a33c0233c5d90c575","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-13 16:29:31.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/26/849626/4/check/dsvm-tempest-cisco-zonemanager-job/a8fe0cb/ : FAILURE in 54m 30s","accounts_in_message":[],"_revision_number":4},{"id":"578a86fe57abb4c3cfb0afabbd832a801b639ae6","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-13 16:32:33.000000000","message":"Patch Set 4:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/26/849626/4/check/fujitsu-eternus-dx-iscsi/234bce4 : SUCCESS in 1h 52m 55s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/26/849626/4/check/fujitsu-eternus-dx-fc/d0c3722 : SUCCESS in 1h 50m 50s","accounts_in_message":[],"_revision_number":4},{"id":"c03fe08d2b4fa8cdc25cbe5cf68d7a9266d9d0ae","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-13 16:42:22.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/26/849626/4/check/tempest-dsvm-full-focal-py3/ce39bc9/ : SUCCESS in 1h 28m 40s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/26/849626/4/check/devstack-plugin-ceph-dsvm-focal-py3/4a85cd0/ : SUCCESS in 1h 41m 08s","accounts_in_message":[],"_revision_number":4},{"id":"89030ae5464f2d336894a8c53206884aa3136539","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-07-13 17:15:13.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/26/849626/4/ : RETRY_LIMIT in 24s","accounts_in_message":[],"_revision_number":4},{"id":"f7d003c57564214f83d7027f9558d230564bf971","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-13 17:20:37.000000000","message":"Patch Set 4:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/15","accounts_in_message":[],"_revision_number":4},{"id":"63dabc6cdbe016bcb8b0a701df6a996859b90bd5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-13 18:00:31.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/26/849626/4/check/iscsi-driver-nimble/458b5ef/458b5ef.html : FAILURE in 1h 43m 26s","accounts_in_message":[],"_revision_number":4},{"id":"84fe214d34cc17c1275aa0a339e57c854a693c92","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-13 18:19:04.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-26-849626-4/ns5_nfs : SUCCESS in 3h 25m 59s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-26-849626-4/ns5_iscsi : SUCCESS in 3h 22m 53s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"591ca59a8368bf9b746b3d8b7f36ecd7ce2e1d54","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-13 18:34:21.000000000","message":"Patch Set 4:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/24","accounts_in_message":[],"_revision_number":4},{"id":"dd8ecd14a2c59f9cb43ef16366cc09b43af17b40","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-13 23:45:24.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/26/849626/4/check/DellEMC_Unity_FC/ca77678/DellEMC_Unity_FC/1152 : SUCCESS in 1h 14m 36s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/26/849626/4/check/DellEMC_Unity_iSCSI/97e752f/DellEMC_Unity_iSCSI/1079 : SUCCESS in 1h 51m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"4c36197e230c9447e67f527754232b5c563059c0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-13 23:51:42.000000000","message":"Patch Set 4:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/42","accounts_in_message":[],"_revision_number":4},{"id":"4336c58ef62156f84bf5a0d1f78825f5a0784955","author":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"date":"2022-07-14 08:32:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"354f7511e86d29460723787ce9f3ee5e38e696bc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 08:38:54.000000000","message":"Patch Set 4:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/53","accounts_in_message":[],"_revision_number":4},{"id":"3fd97d00658bbffa87afe902f5072117a45499fb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-07-14 08:45:55.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b24128ec2ffe4450b1ef8452eb4a28a2 : FAILURE in 11m 54s","accounts_in_message":[],"_revision_number":4},{"id":"eff6a9feeccf136d578fbb5cfee14339f106c911","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 08:56:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cade0c04494c4355b3e33a639b5f9d59\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/22e3adf21f3142768378ffce9130ce7d : SUCCESS in 13m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/abbcf9f93f024ea2b731b444837c08c4 : SUCCESS in 22m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7bd192499a3aa4a5b8d0c8f52ac5910d5098098d","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2022-07-14 09:12:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c5b983503854a48c4dbd3e713272dbed06d03589","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 09:18:55.000000000","message":"Patch Set 4:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/55","accounts_in_message":[],"_revision_number":4},{"id":"8b1d6ce424dfb5848689d727f73f2bee8abbcf81","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-14 09:31:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/849626/4/iscsi/ : SUCCESS in 58m 09s\n- cinder-smb http://cloudbase-ci.com/cinder/849626/4/smb/ : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":4},{"id":"f3a1e671a90426e92f2340c52ebcdc19db8798de","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-14 10:00:38.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/26/849626/4/check/dsvm-tempest-cisco-zonemanager-job/8d271b0/ : FAILURE in 53m 47s","accounts_in_message":[],"_revision_number":4},{"id":"3e34a2d5ce4e31d2e5e006eb5f8918eca19134fe","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-14 10:05:36.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/26/849626/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a5700b7/ : SUCCESS in 1h 32m 01s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/26/849626/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f63f61b/ : SUCCESS in 1h 30m 30s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/26/849626/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/67d3463/ : FAILURE in 28m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a86ca56b573d785b2ed692b6c91ad8f67d6d1f7b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-14 10:26:56.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/26/849626/4/check/DellEMC_PowerStore_iSCSI/92639b4/DellEMC_PowerStore_iSCSI/599 : FAILURE in 38m 06s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/26/849626/4/check/DellEMC_PowerStore_FC/68ce1c9/DellEMC_PowerStore_FC/638 : ABORTED in 1h 14m 08s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"33c868a5e76fb4e7aee16f5fb47f48d87121fbc4","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2022-07-14 10:27:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f0aa823ea0d6750bafbd1b8f5da13f0f73b8e854","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 10:36:42.000000000","message":"Patch Set 4:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/62","accounts_in_message":[],"_revision_number":4},{"id":"dff54730816b6327104d0462116eeba30e8ebfe9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-14 10:42:11.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/26/849626/4/check/tempest-dsvm-full-focal-py3/12ecd64/ : FAILURE in 2h 07m 09s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/26/849626/4/check/devstack-plugin-ceph-dsvm-focal-py3/e063104/ : SUCCESS in 1h 32m 23s","accounts_in_message":[],"_revision_number":4},{"id":"6036e54c5c62551c37da3f03d696bd2965e2893f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 10:50:08.000000000","message":"Patch Set 4:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/64","accounts_in_message":[],"_revision_number":4},{"id":"c8f70069e4b026e85736246e655956bda4a34423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 11:05:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b09f7e36065c4581999a096a8ea31eeb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0db4b7dc146142088683ba36b3d1fba2 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a31a8a491c844629018f3815a643861 : SUCCESS in 11m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d2067ec91ee4815a97d501dc70943cd : SUCCESS in 14m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d565af54058d40e686c65d47d795a941 : SUCCESS in 11m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbe9d82080404fa5acc3922f95652efe : SUCCESS in 16m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/372bbda5819844ae837902a9d12220ee : SUCCESS in 52m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d009d4961f7143fba8b83e068271bf3b : SUCCESS in 1h 26m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b2bdb13123142d3a0b9ef30b7568421 : SUCCESS in 45m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7e6da59bca74dbfb88ce081cd61442e : SUCCESS in 10m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b6615efd774c413ca8accdb17c5656d0 : SUCCESS in 15m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d1319f4ba71043119cd22dd626f9a1b9 : SUCCESS in 7m 38s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7664ff1b7c8b4146ab1c091d3e1141e5 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2fa24ab2b72f40d3a21cf98987302044 : SUCCESS in 9m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/51d616f19d5941eab34a1e8159da256c : SUCCESS in 7m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5c5a811ad9cd4467be3995bee3abacae : SUCCESS in 50m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ad8385847e184bcb8974b30525dade3d : FAILURE in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5a8d5515432d4442885cad288fcd5dfa : SUCCESS in 1h 41m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b1402b47351a4efab23267f898adc563 : FAILURE in 1h 24m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/99f8247ae6b044d9baf4c6710e36bc17 : SUCCESS in 1h 53m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ba78d8ae4874466e81cfa60d479bedef : SUCCESS in 58m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4a8b04c388ca4243b596d8171687de33 : SUCCESS in 1h 31m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c47a0a08a6014d58937d39d820f6fd82 : SUCCESS in 43m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f0c4b0b26d1c4157a49ed79cac3fb6d6 : SUCCESS in 1h 23m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4a2a01d2a169446f989e2dac4cd239cb : SUCCESS in 2h 27m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2759679620b5494bb543a69d79246b98 : SUCCESS in 53m 23s","accounts_in_message":[],"_revision_number":4},{"id":"208cd9dfee6f48f6e9838510f1052d6e94ad8931","author":{"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},"date":"2022-07-14 11:07:36.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"74bf4cfbc9cf0ef35af5d8783d85747733bc91e0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-14 11:17:30.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/849626/4/upstream-check/cinder-tempest-plugin-ontap-fcp/d561cc7/ : RETRY_LIMIT in 10m 16s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/849626/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/0e683f6/ : RETRY_LIMIT in 7m 40s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/849626/4/upstream-check/cinder-tempest-plugin-ontap-nfs/4e4cc0c/ : RETRY_LIMIT in 7m 49s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"e945c14dea3a46daef924f884a8f203337652893","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-14 12:16:01.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/26/849626/4/check/DellEMC_PowerStore_iSCSI/2c67a95/DellEMC_PowerStore_iSCSI/600 : ABORTED in 1h 48m 01s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/26/849626/4/check/DellEMC_PowerStore_FC/ab03e99/DellEMC_PowerStore_FC/639 : FAILURE in 1h 21m 10s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"9e08369c8e627fdb2d941fa5e27db4aa72077dbb","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2022-07-14 12:20:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"45575f2c76ab030d897b29290814bd9e7a711746","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2022-07-14 12:23:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c9429ad2d3b167876a7e26a1eaeaec2bcf4c6a24","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-14 12:29:45.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/26/849626/4/check/DellEMC_PowerFlex/72073a7/DellEMC_PowerFlex/988 : SUCCESS in 1h 24m 05s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"3793cc36668ba1a7b82213ff8a309b376a4225fe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 12:29:46.000000000","message":"Patch Set 4:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":4},{"id":"c88a32e969e48727f46855916365d87ec882b3e1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 13:15:57.000000000","message":"Patch Set 4:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/79","accounts_in_message":[],"_revision_number":4},{"id":"97feb04ca4a95703c99b2a6a58e270678acef71f","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2022-07-15 03:03:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4e670c24029665f4404a58bb6712fca84466c7a9","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2022-07-15 08:44:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c84377f33245cbb3a65457639082501a5d6e0b70","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-15 10:44:54.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/26/849626/4/check/DellEMC_PowerStore_iSCSI/958280d/DellEMC_PowerStore_iSCSI/608 : ABORTED in 2h 00m 02s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/26/849626/4/check/DellEMC_PowerStore_FC/d4fe240/DellEMC_PowerStore_FC/643 : SUCCESS in 1h 20m 37s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"145b69a51ce2da94fe97b665537fef12aebe49de","author":{"_account_id":34290,"name":"Alexander Malashenko","email":"alexander.malashenko@dell.com","username":"amalashenko"},"date":"2022-07-15 11:11:24.000000000","message":"Patch Set 4: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8e83c193e379a004033a13f4b90a6dddc5c18382","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2022-07-15 11:42:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5b9716a5ab643c85aae6d13ebebd6021e896cc5b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-15 13:41:36.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/26/849626/4/check/DellEMC_PowerStore_iSCSI/0639d44/DellEMC_PowerStore_iSCSI/609 : ABORTED in 1h 59m 33s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/26/849626/4/check/DellEMC_PowerStore_FC/b8351ac/DellEMC_PowerStore_FC/644 : SUCCESS in 1h 20m 23s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"eb65ef203a8363ddb977bdb8da67984e760c666e","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-07-15 14:17:20.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f703f4dbdd22e75442bf86a5c58587b838ab4d4d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-07-15 14:28:19.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fb5baf8f6abb0689366bc13144ab32e70cbba153","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 14:29:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0198a7041bcd05494d25d3dd5738084cbe2a12f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 18:24:36.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7757887376804ce58518a75395351748\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a15a11ab33942228d4e9b8543782978 : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87c54fbed4d348c39b7fbc8cf78201fc : SUCCESS in 12m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a9c933eb54b4b8b9ddccb85d483db98 : SUCCESS in 12m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c076c5c499864838a0fcbbc4823a160b : SUCCESS in 13m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/1ef481197b684c238d614862b4daf227 : SUCCESS in 1h 00m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fd2c39d556c34ae7842ebaf12dd2b858 : SUCCESS in 1h 29m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7e59bd8ce23493ca8b0ec618b3dc6be : SUCCESS in 56m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c654ba9f873410eb2911948a57cfb85 : SUCCESS in 10m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9f5fd62df5984608ac2ff810f798484b : SUCCESS in 1h 28m 04s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f266f7e5aafd424894deb7a747927e2a : SUCCESS in 1h 48m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d1862426f0f4859940d6d2fabc0db50 : SUCCESS in 51m 40s","accounts_in_message":[],"_revision_number":4},{"id":"6015a97d3f1508c5e654ea006e39ae5e7c888781","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 18:24:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"8fbe9a2285a0e580ebb1593fba18b671344cc85b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 18:26:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b58d6ff550d2449dae38de51954a8f7d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bb86791b7af43bc8be00cdb5e263711 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dbeeb6f37654953bdd291c2898551d4 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":4},{"id":"e57486c1795f97e577cf69f4718854de67bac6e3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-18 01:19:30.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"f512ef533bd869b3d4d472fd9af50f9007600131","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-18 01:19:43.000000000","message":"Patch Set 2:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":2},{"id":"0e83cbc714cdce5c0140593ea239d43fcece76fc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-18 01:19:47.000000000","message":"Patch Set 3:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":3},{"id":"1e76096e270a10a608c91bb9df0b810477d0210c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-18 01:20:31.000000000","message":"Patch Set 4:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a4ab9220fcfd4afd3019a0f3bd4f4b452296166f","revisions":{"c1ab3ca5d37c1e18ad11d775aefa5e53397bc619":{"kind":"REWORK","_number":1,"created":"2022-07-13 08:08:41.000000000","uploader":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"ref":"refs/changes/26/849626/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/26/849626/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/849626/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/849626/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/849626/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/26/849626/1"}}},"commit":{"parents":[{"commit":"a2deecef891a16703e5410eb7805442961307090","subject":"Merge \"mypy: service.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2deecef891a16703e5410eb7805442961307090"}]}],"author":{"name":"olegnest","email":"Oleg_Nesterenkov@Dell.com","date":"2022-07-13 08:06:07.000000000","tz":180},"committer":{"name":"olegnest","email":"Oleg_Nesterenkov@Dell.com","date":"2022-07-13 08:06:07.000000000","tz":180},"subject":"PowerStore driver - Request data validation fix","message":"PowerStore driver - Request data validation fix\n\nCloses-Bug: #1981068\nChange-Id: I0b389f4f53d3b34eddc23a7aebc85a5b42f3a09f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1ab3ca5d37c1e18ad11d775aefa5e53397bc619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1ab3ca5d37c1e18ad11d775aefa5e53397bc619"}]},"branch":"refs/heads/master"},"515532c2dbfb11abb14a4006d853bcfd9588781c":{"kind":"REWORK","_number":2,"created":"2022-07-13 08:12:58.000000000","uploader":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"ref":"refs/changes/26/849626/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/26/849626/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/849626/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/849626/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/849626/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/26/849626/2"}}},"commit":{"parents":[{"commit":"a2deecef891a16703e5410eb7805442961307090","subject":"Merge \"mypy: service.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2deecef891a16703e5410eb7805442961307090"}]}],"author":{"name":"olegnest","email":"Oleg_Nesterenkov@Dell.com","date":"2022-07-13 08:06:07.000000000","tz":180},"committer":{"name":"olegnest","email":"Oleg_Nesterenkov@Dell.com","date":"2022-07-13 08:12:49.000000000","tz":180},"subject":"PowerStore driver - Request data validation fix","message":"PowerStore driver - Request data validation fix\n\nCloses-Bug: #1981068\nChange-Id: I0b389f4f53d3b34eddc23a7aebc85a5b42f3a09f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/515532c2dbfb11abb14a4006d853bcfd9588781c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/515532c2dbfb11abb14a4006d853bcfd9588781c"}]},"branch":"refs/heads/master"},"bd58c60b9d2a74dc8b76da1af6e4f6ac28df1d4a":{"kind":"REWORK","_number":3,"created":"2022-07-13 08:14:43.000000000","uploader":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"ref":"refs/changes/26/849626/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/26/849626/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/849626/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/849626/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/849626/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/26/849626/3"}}},"commit":{"parents":[{"commit":"a2deecef891a16703e5410eb7805442961307090","subject":"Merge \"mypy: service.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2deecef891a16703e5410eb7805442961307090"}]}],"author":{"name":"olegnest","email":"Oleg_Nesterenkov@Dell.com","date":"2022-07-13 08:06:07.000000000","tz":180},"committer":{"name":"olegnest","email":"Oleg_Nesterenkov@Dell.com","date":"2022-07-13 08:14:34.000000000","tz":180},"subject":"PowerStore driver - Request data validation fix","message":"PowerStore driver - Request data validation fix\n\nCloses-Bug: #1981068\nChange-Id: I0b389f4f53d3b34eddc23a7aebc85a5b42f3a09f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd58c60b9d2a74dc8b76da1af6e4f6ac28df1d4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd58c60b9d2a74dc8b76da1af6e4f6ac28df1d4a"}]},"branch":"refs/heads/master"},"a4ab9220fcfd4afd3019a0f3bd4f4b452296166f":{"kind":"REWORK","_number":4,"created":"2022-07-13 14:38:59.000000000","uploader":{"_account_id":33502,"name":"Oleg","display_name":"Oleg Nesterenkov","email":"Oleg_Nesterenkov@Dell.com","username":"olegnest"},"ref":"refs/changes/26/849626/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/26/849626/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/849626/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/849626/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/849626/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/26/849626/4"}}},"commit":{"parents":[{"commit":"a2deecef891a16703e5410eb7805442961307090","subject":"Merge \"mypy: service.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2deecef891a16703e5410eb7805442961307090"}]}],"author":{"name":"olegnest","email":"Oleg_Nesterenkov@Dell.com","date":"2022-07-13 08:06:07.000000000","tz":180},"committer":{"name":"olegnest","email":"Oleg_Nesterenkov@Dell.com","date":"2022-07-13 14:33:51.000000000","tz":180},"subject":"PowerStore driver - Request data validation fix","message":"PowerStore driver - Request data validation fix\n\nDue to json.dumps({}) method returns {}, an error occurs in some\nclient\u0027s methods, such as when deleting volumes. This fix validate\nif any payload was provided for the request.\n\nCloses-Bug: #1981068\nChange-Id: I0b389f4f53d3b34eddc23a7aebc85a5b42f3a09f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a4ab9220fcfd4afd3019a0f3bd4f4b452296166f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a4ab9220fcfd4afd3019a0f3bd4f4b452296166f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
